)]}'
{"id":"openstack%2Fcinder~775931","triplet_id":"openstack%2Fcinder~master~I680cb8c4be13bcd3ac6785a7afa81ce5d3477f91","project":"openstack/cinder","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I680cb8c4be13bcd3ac6785a7afa81ce5d3477f91","subject":"Properly handle InvalidScope exceptions","status":"MERGED","created":"2021-02-16 20:49:57.000000000","updated":"2021-03-05 23:52:20.000000000","submitted":"2021-03-05 05:29:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775931-secure-rbac","meta_rev_id":"001f764d875e204e827913bc004c3a4187cca8d7","_number":775931,"virtual_id_number":775931,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 05:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-02 22:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-03 13:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-03 13:42:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 21:29:57.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 21:38:05.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-16 21:51:49.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 22:04:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 23:05:09.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-02-16 23:15:07.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-02-17 01:09:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 01:29:21.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 01:59:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 02:05:41.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-17 03:50:54.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 05:09:08.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-17 06:14:37.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-17 13:55:29.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-17 15:48:31.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-02-17 15:49:00.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 16:43:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-17 17:49:20.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-18 00:11:31.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-02-18 21:14:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-02-19 08:12:37.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 08:42:43.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-20 08:29:09.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-02-20 11:06:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-21 21:00:44.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-02 22:35:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-03 13:42:16.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"31536dce6ed0cb813a14db0f3dcfd5ca05a8d252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-16 20:49:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7734f5bf6bde2fb9d39298fc2812936c848ea8a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-16 21:29:57.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/1/check/tempest-dsvm-full-focal-py3/a469fff/ : FAILURE in 32m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/1/check/devstack-plugin-ceph-dsvm-focal-py3/b10d9b6/ : FAILURE in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"48d634fc93191022be59bf9ecb3bdb5bc8bcf030","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-16 21:38:05.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775931/1/check/dsvm-tempest-TOYOU-ACS5000-driver/2ec75b6 : FAILURE in 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e58c68a0ee3f92bdfefa8081ff53e87861c745f4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-16 21:38:59.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"b6d64fff6bf7d6c6224674039d178a8fcf5d51fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-16 21:51:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcf390cb2f9e2370f4d95af69a9f3a50cf2e9be2","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-16 21:51:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad11c84c9fba4bb550eeb9be7f93f1c38f039c24","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-16 22:04:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/775931/2/check/iscsi-driver-primera/6225bbd/6225bbd.html : FAILURE in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e2cfb70a38ccf0b34ff4eee1ba20cd11e700fbee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-16 22:07:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/775931/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bd71da1/bd71da1.html : FAILURE in 9m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7c41e2a4eb7424c30de0941edeedbe6b08d7995e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-16 22:23:21.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775931/2/check/dsvm-tempest-TOYOU-ACS5000-driver/66afe12 : FAILURE in 22m 33s","accounts_in_message":[],"_revision_number":2},{"id":"44312d0d5bfc65195812ac42d2063b5406749440","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-16 22:39:44.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/2/check/tempest-dsvm-full-focal-py3/c2ac5ce/ : FAILURE in 43m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4c6bfe66567847e5ab0428ead83ba1ae : POST_FAILURE in 36m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b799b45a50626cf51476d327709ed545a88ddea7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-16 23:05:09.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775931/2/iscsi/ : SUCCESS in 1h 12m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/775931/2/smb/ : FAILURE in 46m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5dc7ea63581f70422cba405c0cb6526c09aea906","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-16 23:15:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775931/2/check/dsvm-tempest-cisco-zonemanager-job/48ed878/ : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d9164c7c5cc5a0899ecf8be3039605fc3d8b7abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 01:09:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99aa6adb76384f9091691fa37b4d824e : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512895d0ff0b424e8b4f6c405dbc0672 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4de6c08d29104aaaa8fe6130139a4eca : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6bdfb7911754cd4822e0d68b1d7b8c5 : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fca13048fda43cd9a7090e1615c7e61 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89554eefa9f404caef8f05f4d91890c : SUCCESS in 11m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/604f6152b15e4fdeb52a534c6c216d47 : SUCCESS in 2m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/ac3fc0b1ffd0490e889e40f800bf5b50 : SUCCESS in 1h 08m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/657acacf30074819a7e680d38a1b4474 : SUCCESS in 1h 17m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d004042826a43bdae9d72ea9fe521ef : SUCCESS in 13m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2426ee44f0b544cb81645f56757680f6 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b4f2d4d86824620b3b7b75ede760d6b : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9cb8dd9e2cf542cb86b4083a138308f4 : SUCCESS in 6m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51587ef14b5a44688673e5125fda9435 : SUCCESS in 31m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aee29b1676d449d4aa6aab63ce956bcf : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0242cd5d854147ea8faae23169dca1e9 : SUCCESS in 1h 08m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d553100c13e444e0a00f4c989d90101b : FAILURE in 1h 23m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97dc8d03d56246e9a50309590cd3d803 : SUCCESS in 1h 32m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d9d13e7131ae46b89b193ff5221b9275 : SUCCESS in 1h 15m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25faab33c0fd42f990928da6fbed9157 : FAILURE in 1h 23m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/234a1c778e654b83b6119e0814d7c82d : SUCCESS in 1h 33m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f1fbac0c06c4bf9b8ecbbbf5358b1ac : SUCCESS in 2h 04m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1623e6574b724c0da03af5cd4ade6f3d : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"2eb01743771471adbcab4229f004e3e5294591a9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-17 01:29:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/775931/2/ : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9ba4db27736ce2cb9800e2f6d72463be4c8cd4e4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-17 01:59:22.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-1/ns5_nfs : FAILURE in 4h 37m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-1/ns5_iscsi : FAILURE in 2h 54m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8edc0dd67b6beee37f8d2fd37da052883923b656","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-17 02:05:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/31/775931/2/upstream-solidfire-check/cinder-solidfire-iSCSI/82429af/ : SUCCESS in 54m 35s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"900c0b3ff19e7fdf8df683e7268b1210c6db264d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-17 03:25:44.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-2/ns5_nfs : FAILURE in 5h 33m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-2/ns5_iscsi : SUCCESS in 5h 22m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"15d9d1ff278e961e6e261dbb4265698bd4616909","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-17 03:50:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/31/775931/2/upstream-check/cinder-cDOT-iSCSI/3d0d8bf/ : SUCCESS in 37m 37s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/31/775931/2/upstream-check/cinder-cDOT-iSCSI-IPv6/3d4d572/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/31/775931/2/upstream-check/cinder-cDOT-NFS/9d8ab37/ : SUCCESS in 37m 32s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/31/775931/2/upstream-check/cinder-cDOT-NFS-IPv6/15aa635/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/31/775931/2/upstream-check/cinder-cDOT-FCP/478dbd6/ : FAILURE in 2h 04m 26s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0afac726abeca7aebc433837bbc4663da4010421","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-17 05:09:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/775931/2/check/DellEMC_PowerStore_iSCSI/d24cd0d/DellEMC_PowerStore_iSCSI/416 : SUCCESS in 53m 53s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/775931/2/check/DellEMC_PowerStore_FC/8dac867/DellEMC_PowerStore_FC/410 : SUCCESS in 46m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"0338846859dc494d3697154443e05359db94ad63","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-17 06:14:37.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775931-2/PVME-iSCSI : FAILURE in 15:18.95\n* PVME-FC http://os-logs.tristero.net/dell-ci/775931-2/PVME-FC : FAILURE in 15:03.99","accounts_in_message":[],"_revision_number":2},{"id":"9dbf35a56418dbb4314721450e97d06be6a0bf5b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-17 13:55:29.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/775931-2/Lenovo-iSCSI : SUCCESS in 48:20.30\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/775931-2/Lenovo-FC : SUCCESS in 52:16.37","accounts_in_message":[],"_revision_number":2},{"id":"6be0fba2490e028be4e9cabcde6bb823a6616b87","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-17 15:48:31.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/31/775931/2/check/DellEMC_SC_FC/d1b5f05/DellEMC_SC_FC/343 : FAILURE in 2h 03m 26s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/31/775931/2/check/DellEMC_SC_iSCSI/f8229f3/DellEMC_SC_iSCSI/342 : FAILURE in 2h 06m 53s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"8b3b974adb512e61273bb3c040aa588fb2e1fb41","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-17 15:49:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775931/2/check/DellEMC_XtremIO_FC/527817e/DellEMC_XtremIO_FC/775 : SUCCESS in 2h 04m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775931/2/check/DellEMC_XtremIO_iSCSI/ff65bcd/DellEMC_XtremIO_iSCSI/777 : SUCCESS in 1h 40m 25s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"40f979574d844186a71df865cadeedb0dda96607","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-17 16:43:29.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/775931/2/check/DellEMC_Unity_FC/dd27644/DellEMC_Unity_FC/714 : SUCCESS in 2h 00m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/775931/2/check/DellEMC_Unity_iSCSI/cd3c359/DellEMC_Unity_iSCSI/675 : SUCCESS in 1h 55m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"70037a2e35169705c71150b487b37fe75ac92344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-17 17:48:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6e103c2a610c6ab23ce45dbb16a357cf971f2489","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-17 17:49:20.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775931-2/HPMSA-iSCSI : FAILURE in 1:05:12\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775931-2/HPMSA-FC : SUCCESS in 34:43.34","accounts_in_message":[],"_revision_number":2},{"id":"b72ed8ab73fca89450871a72d8795f96341209e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 18:28:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/775931/3/check/iscsi-driver-primera/6703281/6703281.html : FAILURE in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"8ca5a6ec03a5f52f752d7115f87b70f3e55504bf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 18:38:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/775931/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/abb53cc/abb53cc.html : FAILURE in 17m 28s","accounts_in_message":[],"_revision_number":3},{"id":"0efffa6639782e556865bf670ebc45748af45c87","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-17 19:23:03.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775931/3/check/dsvm-tempest-TOYOU-ACS5000-driver/94bf050 : FAILURE in 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"1f0ada28ea042ae4f0fede3ad1d6555cdef502bf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 19:42:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775931/3/iscsi/ : SUCCESS in 1h 28m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/775931/3/smb/ : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":3},{"id":"2c037f619fb1ee06aee4cc64038a1fb24e95614e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:46:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779e054436d24fb592eea669fce7196b : FAILURE in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d23113a62785448a83496c93a2ac110a : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26692e49a9594270ab8446b800588a41 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65040ac0838b46489fbf7757489e5212 : SUCCESS in 9m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50ee14816dad4bb49163827fc5f1e3fa : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f4aa2abccc4ccc9ce796bc4a1dabb5 : SUCCESS in 11m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/769933cc97904eeb8c1ee777ccd421d8 : SUCCESS in 4m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/9872e64bd8a04837afe8c301b4ab0bf0 : SUCCESS in 1h 20m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1b05b4bf5eb47388542f01f2302b437 : SUCCESS in 1h 19m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1c5570dd4aef4c2faabdca871ca4bc45 : SUCCESS in 14m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de3bf2f7a3e342e397008accccd4d454 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1875696ef5c74035a5771ed31161147b : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ff27863ea374521b06b522ae2af4a4f : SUCCESS in 7m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2319dead679426a923d3970df66894c : SUCCESS in 50m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/796ae1268eb44e12ab316ab2e2f9ab11 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97a8ec5bdb724de1b817813ed3257099 : SUCCESS in 1h 16m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4b07c2385256434b9a74c4f2efb46aee : FAILURE in 1h 35m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef547a5d224a486a9349d66f70fb79a6 : SUCCESS in 1h 43m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e8c09fdd5cda450cad2bb6ee49f87be6 : SUCCESS in 1h 21m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/328c911fb2ea4c72b253c0fb7a7ddfa1 : SUCCESS in 1h 23m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/561df4f5953c4eaeae6ab066f1563bdc : SUCCESS in 1h 25m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/99cee8efcfc54f6594f66b7388cc40cd : SUCCESS in 2h 06m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c9227ab8ce74e1ba1a79048855c30cf : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":3},{"id":"ec5c981eac509e8b7bc7aaa5f0d2c1a2bec44bbb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-17 21:00:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775931/3/check/dsvm-tempest-cisco-zonemanager-job/c0489e3/ : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7726cb393c4ee66a4c7a1233ece4a988c90a7f93","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-17 22:42:33.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/3/check/tempest-dsvm-full-focal-py3/626ff6e/ : SUCCESS in 2h 07m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/3/check/devstack-plugin-ceph-dsvm-focal-py3/1f815e7/ : SUCCESS in 1h 40m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9e088df923a3f864b3f672ac5875d52000375763","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-18 01:34:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_nfs : SUCCESS in 5h 22m 46s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_iscsi : SUCCESS in 5h 12m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"ac6b4dc50f22313cad7f3c42d03c3477d0a75c93","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-18 10:37:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/775931/3/ : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":3},{"id":"d36843d4da3758e8627d077fa3c77b2d40b47775","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 15:14:06.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_XtremIO_FC/16a2db6/DellEMC_XtremIO_FC/789 : SUCCESS in 2h 04m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_XtremIO_iSCSI/3be9a49/DellEMC_XtremIO_iSCSI/791 : SUCCESS in 1h 41m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"74b16350b0c0af6f506bd03a419b5b2a6bf37c13","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-18 17:27:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_Unity_FC/123364d/DellEMC_Unity_FC/727 : FAILURE in 1h 50m 18s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_Unity_iSCSI/6a6d091/DellEMC_Unity_iSCSI/689 : FAILURE in 1h 58m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"818857b1a4929de4561da1702154574b8fd4b73b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-18 18:10:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e2c703215e6ddce3e9aa6e843876eb0f192dc52e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-18 19:45:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775931/3/check/dsvm-tempest-cisco-zonemanager-job/11d0103/ : SUCCESS in 1h 20m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e8639b46d0859599a703727d713eab23bdd37866","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-18 19:56:11.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775931/3/iscsi/ : FAILURE in 1h 45m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/775931/3/smb/ : FAILURE in 1h 06m 28s","accounts_in_message":[],"_revision_number":3},{"id":"a9b95a9da6a32b2d7f6cd62afb1528af07358be0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-18 21:14:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/775931/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/eea2dcd/ : TIMED_OUT in 3h 03m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/775931/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f10e8a5/ : SUCCESS in 1h 51m 42s","accounts_in_message":[],"_revision_number":3},{"id":"bb5a4281343e3e060715cedc87f1d780aa36659e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 01:36:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b218144fe2cf416c94e50d5ed7fe41e0 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a1134dc15344afa13aee766b931c3b : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/519725a00f20491998b3ec603600535c : SUCCESS in 10m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/972bda68aa9842949d4e90a824463408 : SUCCESS in 12m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5297c039ae24e8c86ef25b573c61187 : SUCCESS in 12m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23eadb6accc849f18e6606bfc5803992 : SUCCESS in 12m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/38df53493fcc40a683b2b2c5a10cb296 : SUCCESS in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/71018d10f71242c2baff40beb37cb477 : SUCCESS in 1h 09m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfdec934978d441086142dc038c0b6f4 : SUCCESS in 1h 18m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fabb3404012f4823af8f95d1715e9220 : SUCCESS in 14m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/075ddc8bfc0840098b9c7bbdb6d700f7 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54c437f08d924e2fa246c2883b7b7a99 : SUCCESS in 7m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92d2095d92384197a0e92e745a21d562 : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/38083556cd8f4c3ba6950a3e4ef900fd : SUCCESS in 44m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c44d12ec70464fe883edc7e92899288a : SUCCESS in 8m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88626dd5d7254179a138e36ec5a98294 : SUCCESS in 59m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/07c3254be2db49239118a80c13aa5ca3 : FAILURE in 1h 49m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ace7f425255649c0bb313631d73c4a11 : SUCCESS in 1h 51m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/024e47cfafcf437f9ed34b5ccb524181 : SUCCESS in 1h 10m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/348959d6f8d8465999cf72e154e090dc : SUCCESS in 1h 14m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e64ba80b73e04993a28f855baa5de212 : SUCCESS in 1h 17m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7935605b574f42ecb20fc34e37344ec0 : SUCCESS in 2h 00m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7be340ba6ce4a8686ffdce195bc42c0 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ceb9940ee05f1858b832c0f31ecec4481fbbf330","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-19 01:46:17.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/3/check/tempest-dsvm-full-focal-py3/3f05b99/ : SUCCESS in 2h 03m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/775931/3/check/devstack-plugin-ceph-dsvm-focal-py3/c57d793/ : SUCCESS in 2h 19m 45s","accounts_in_message":[],"_revision_number":3},{"id":"5802beb366b594b38f022e6fb95cc4f09d7509fc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-19 02:33:42.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerStore_iSCSI/95c98a7/DellEMC_PowerStore_iSCSI/445 : SUCCESS in 57m 14s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerStore_FC/0878a45/DellEMC_PowerStore_FC/439 : SUCCESS in 50m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"ea294ff5371184e5149219e3c21a539af4e799f0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-19 02:56:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/31/775931/3/upstream-solidfire-check/cinder-solidfire-iSCSI/d70ae3c/ : SUCCESS in 1h 19m 31s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"fc11183d44d509d895420c8ad9379805b8aa9873","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-19 04:42:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-iSCSI/127377f/ : SUCCESS in 36m 33s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-iSCSI-IPv6/b5b9990/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-NFS/4b968f4/ : SUCCESS in 2h 14m 02s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-NFS-IPv6/13a3c44/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-FCP/0e8b431/ : SUCCESS in 57m 53s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"8afba9b640dfa7c6b9190fb18bfc436775f9855d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-19 04:57:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_SC_FC/ca13800/DellEMC_SC_FC/362 : FAILURE in 1h 57m 40s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_SC_iSCSI/44dd15d/DellEMC_SC_iSCSI/360 : FAILURE in 2h 05m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"d12119ef71f3f40f4f02ce6f59e1677071363ed3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 08:12:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_VNX_iSCSI/14b3658/DellEMC_VNX_iSCSI/415 : FAILURE in 1h 21m 05s\n- DellEMC_VNX_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_VNX_FC/09cc669/DellEMC_VNX_FC/434 : FAILURE in 52m 54s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"2909171d28eba693a9a06e44e9737ccde3c18598","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 08:42:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerFlex/083cb84/DellEMC_PowerFlex/413 : FAILURE in 2h 45m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"c9f57c20723e0362012081490e6af3fc21b958bc","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-19 19:48:45.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775931-3/PVME-iSCSI : SUCCESS in 47:48.45\n* PVME-FC http://os-logs.tristero.net/dell-ci/775931-3/PVME-FC : SUCCESS in 48:06.30","accounts_in_message":[],"_revision_number":3},{"id":"98d0426954ff29a33d559f16d41f88ed3a4540a4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-20 05:33:39.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_nfs : FAILURE in 6h 56m 13s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_iscsi : FAILURE in 4h 13m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"70029df00aa2081bcf612531287365cd9b4fb4a7","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-20 08:29:09.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/775931-3/Seagate-iSCSI : SUCCESS in 43:55.59\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/775931-3/Seagate-FC : SUCCESS in 39:55.11","accounts_in_message":[],"_revision_number":3},{"id":"07a28d4bcbf910c18c998305851f17b6e58d35d1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 11:06:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775931/3/20210220174759/NEC-ISCSI : SUCCESS in 2h 18m 15s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775931/3/20210220174759/NEC-FC : SUCCESS in 2h 18m 32s","accounts_in_message":[],"_revision_number":3},{"id":"1ba28cc5cdab9c27c89f6e156eb19d337f1f9b2a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 11:19:05.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_XtremIO_FC/fcb21f7/DellEMC_XtremIO_FC/810 : SUCCESS in 1h 57m 12s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_XtremIO_iSCSI/ca68ce4/DellEMC_XtremIO_iSCSI/812 : SUCCESS in 1h 40m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"447e319bca7a41bbdc7167d27445e29882db72b3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-20 12:15:34.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/775931-3/Lenovo-iSCSI : SUCCESS in 46:46.49\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/775931-3/Lenovo-FC : SUCCESS in 43:49.76","accounts_in_message":[],"_revision_number":3},{"id":"e9974f92c08de1b61076e8b5f7fa19b08424a3b3","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-21 00:09:42.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775931-3/HPMSA-iSCSI : SUCCESS in 59:14.06\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775931-3/HPMSA-FC : SUCCESS in 35:04.94","accounts_in_message":[],"_revision_number":3},{"id":"ebf980a59d51d104e2aa6a337580587602cad482","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-21 21:00:44.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerMAX_FC/b3ec69c/DellEMC_PowerMAX_FC/484 : SUCCESS in 5h 06m 29s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerMAX_iSCSI/6799747/DellEMC_PowerMAX_iSCSI/448 : SUCCESS in 4h 44m 17s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"cd6dc4d9e23a441b33c4161c22f53adb9e737698","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-22 18:45:31.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_nfs : SUCCESS in 5h 21m 7s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_iscsi : FAILURE in 3h 20m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a47d6b946c81d2b58ab6f5717cabd058f1613688","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-23 18:15:16.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_nfs : SUCCESS in 4h 11m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775931-3/ns5_iscsi : SUCCESS in 4h 14m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"e1438122c7d60482e79269120b7043ed19e5006e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 22:35:56.000000000","message":"Patch Set 3: Code-Review+2\n\nThis looks like the right way to handle this.","accounts_in_message":[],"_revision_number":3},{"id":"592d91009aacf1cfbfefcdcd8beef691ec00a5f7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-03 13:42:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThis change looks good to me.  Thanks Lance.","accounts_in_message":[],"_revision_number":3},{"id":"4221b23233540ad91f096cf49f6af953f05830a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:43:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7f113fa95484dedf2eb3a22f1f19d6ba3dc6236d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:18:01.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"b5ba5a6c6ab87fedd5f6812ed415358f57ebad6b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-04 03:04:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c3b623227440bdd5000f1c5ef753544c46aed77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 05:14:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/485ff10ae0cd463da59291239b140f90 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb3c05fbeb145aa9d9ef73af1411699 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2e9dcc8d94d4a7093117ee7ee36b0c6 : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/163c99513e824cfda87c508dfc325605 : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5e4175090d046e09f71b13ff929ab31 : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6535403d284d6c9be087daeaaf924b : SUCCESS in 11m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ff2bc38434a49ecb207415265bd10c0 : SUCCESS in 4m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/242b0bff2f2f4e64af1474cf527f3f03 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c45d9f2dd2b4a778c6da8c64fcdc1da : SUCCESS in 1h 48m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7bf3760d773243a0a25aa5d63112b8da : SUCCESS in 12m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/581b72bb4ecc44ac8441588ca75fbb07 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f507b391a9c464c97ebce6707b9c380 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/09b184f9dd84466ab3ba293c52b0b13e : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28b02994ad684cbc92ae99c8c1cf8b38 : SUCCESS in 39m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55c0e6dcb3b84affab4a8a8e1f04e4ba : SUCCESS in 9m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/42357c0709a4460d855519a920d7fc96 : SUCCESS in 1h 00m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d706f1e69b044e63bc537915f49ac90f : FAILURE in 1h 22m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0340bcdbdaa3453c9f59029e545f6e91 : SUCCESS in 1h 19m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/47b73a15fbef4aa1ad7a851d714b1f0f : SUCCESS in 1h 07m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c237c95d7c304927a3d95d640ea8c1d4 : FAILURE in 1h 51m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/69b20053b23d465a8da104bc55daf7f9 : SUCCESS in 1h 56m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02e099de4f3b457087c5a4a9eb0f7fa8 : SUCCESS in 1h 53m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18be9675168c4a4ab61d8574198139ad : SUCCESS in 1h 25m 41s","accounts_in_message":[],"_revision_number":3},{"id":"cab15ac03e8d72b92e623aa1ab52f13b7304298f","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-04 07:15:58.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775931-3/PVME-iSCSI : FAILURE in 15:07.31\n* PVME-FC http://os-logs.tristero.net/dell-ci/775931-3/PVME-FC : FAILURE in 15:10.92","accounts_in_message":[],"_revision_number":3},{"id":"42e258a2904c8210e5b7edcfa449bdc1d18d0d20","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-04 07:16:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/775931/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7c9cc8a/ : SUCCESS in 2h 18m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/775931/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/44c0777/ : SUCCESS in 1h 58m 33s","accounts_in_message":[],"_revision_number":3},{"id":"77bd36e66b7c5e13ad455a7330aa9c03dac81938","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 07:49:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775931/3/check/dsvm-tempest-cisco-zonemanager-job/812ef52/ : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6f8cba1a9a11877eb82bde0b094154cd0f6a9327","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-04 08:23:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/31/775931/3/upstream-solidfire-check/cinder-solidfire-iSCSI/1548741/ : FAILURE in 56m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"d1e98622a2559beed9201951dafe556929ffb312","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-04 10:36:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-iSCSI/45f03fa/ : SUCCESS in 42m 09s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-iSCSI-IPv6/6da03f7/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-NFS/17d1b41/ : SUCCESS in 39m 02s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-NFS-IPv6/82a5ee9/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/31/775931/3/upstream-check/cinder-cDOT-FCP/1254ef3/ : SUCCESS in 1h 37m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c3b0f2d7614ca03061d7bfaa35e36f9a72eba2e6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-04 10:38:21.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerStore_iSCSI/6a21c8a/DellEMC_PowerStore_iSCSI/503 : SUCCESS in 53m 30s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerStore_FC/926f5ba/DellEMC_PowerStore_FC/497 : SUCCESS in 49m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"cdd96529bacb96254f5eb7cc886acec4167affc6","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-04 15:13:36.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/775931-3/Seagate-iSCSI : SUCCESS in 46:35.35\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/775931-3/Seagate-FC : SUCCESS in 46:30.58","accounts_in_message":[],"_revision_number":3},{"id":"76af4ca11a6fdbacac51968432ab20305eaf41f5","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-04 16:14:10.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775931-3/HPMSA-iSCSI : FAILURE in 1:09:32\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775931-3/HPMSA-FC : SUCCESS in 35:21.50","accounts_in_message":[],"_revision_number":3},{"id":"33a8a5b17aa5239d5eccda81da667bd1045b2120","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 22:44:42.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2a543743b0f94e85bd68d83856022f13 : POST_FAILURE in 1h 49m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2249e2dc946a4b0b91b578379faaac46 : POST_FAILURE in 2h 24m 49s","accounts_in_message":[],"_revision_number":3},{"id":"2d12fb8fb1341a8b26ffc3adb35a45c7c1c4fdb5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 00:40:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerMAX_FC/c73e68d/DellEMC_PowerMAX_FC/527 : SUCCESS in 5h 43m 49s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerMAX_iSCSI/1efec9a/DellEMC_PowerMAX_iSCSI/492 : FAILURE in 4h 34m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"921eff71535d7d9201b8153529e938751aa7a89d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:58:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1dde87eaad618918ef2a2edf0adc8a464327ccf6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 04:00:16.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/31/775931/3/check/DellEMC_PowerFlex/c1633c0/DellEMC_PowerFlex/477 : SUCCESS in 2h 40m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"6d0ef10bc36ecbab23b3bd90bd66573589715031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 05:29:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c1ada9cc6014991aa761ac1f880b06b : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654b55e3f1e84917aa6cbf81cf48d4e9 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36536d3d1cbe41c79ffa329c8f556a90 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/557790f9716549d79f07656956dce512 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006c76fcbe324b27bea57291f2ee1ecb : SUCCESS in 11m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d25439222a7b43bea17004913adb5628 : SUCCESS in 3m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/23a1a976f3fd4eb3aee095092672140b : SUCCESS in 1h 00m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d66542934d04d3a8298c4c381604296 : SUCCESS in 1h 06m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/48811c9b29134b159ace9abb4e9bec31 : SUCCESS in 1h 07m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efe66bbb126148cf9ab643014cc99a93 : SUCCESS in 1h 03m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ad55e6d349c42aaaed08a093c96bea8 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":3},{"id":"46943114867854ec7324855ae4271c137979f3df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 05:29:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7765c4202627280cb17e087c1c3f35464ac49cef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 05:32:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a10b961d1a64eb5b97699866227ae3f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3},{"id":"001f764d875e204e827913bc004c3a4187cca8d7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 23:52:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/31/775931/3/check/DellEMC_VNX_iSCSI/e7744b0/DellEMC_VNX_iSCSI/485 : SUCCESS in 3h 45m 04s\n- DellEMC_VNX_FC http://publiclogs.emc.com/31/775931/3/check/DellEMC_VNX_FC/094ac14/DellEMC_VNX_FC/505 : SUCCESS in 1h 20m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"118da903b1ff15805ae2efaf397d139145e1cf0c","revisions":{"7fc8a1fc2d11807f165ed640dec6077be70d0663":{"kind":"REWORK","_number":1,"created":"2021-02-16 20:49:57.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/31/775931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775931/1"}}},"commit":{"parents":[{"commit":"889611b531b37532ad6229ff85f150ae38803832","subject":"Update secure RBAC check strings with descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/889611b531b37532ad6229ff85f150ae38803832"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-16 19:44:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-16 20:48:37.000000000","tz":0},"subject":"Properly handle InvalidScope exceptions","message":"Properly handle InvalidScope exceptions\n\nNow that we\u0027re starting to set scope_types on default policies, we\nshould make sure we handle InvalidScope exceptions from oslo.policy in\nthe event enforce_scope\u003dTrue. Operators won\u0027t use this switch for a\nwhile, but it prepares us for when cinder will be system-scope aware.\n\nChange-Id: I680cb8c4be13bcd3ac6785a7afa81ce5d3477f91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fc8a1fc2d11807f165ed640dec6077be70d0663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fc8a1fc2d11807f165ed640dec6077be70d0663"}]},"branch":"refs/heads/master"},"7aea7628b3def5efc8ef2b08d84257c099e0c55b":{"kind":"REWORK","_number":2,"created":"2021-02-16 21:51:09.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/31/775931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775931/2"}}},"commit":{"parents":[{"commit":"889611b531b37532ad6229ff85f150ae38803832","subject":"Update secure RBAC check strings with descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/889611b531b37532ad6229ff85f150ae38803832"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-16 19:44:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-16 21:48:22.000000000","tz":0},"subject":"Properly handle InvalidScope exceptions","message":"Properly handle InvalidScope exceptions\n\nNow that we\u0027re starting to set scope_types on default policies, we\nshould make sure we handle InvalidScope exceptions from oslo.policy in\nthe event enforce_scope\u003dTrue. Operators won\u0027t use this switch for a\nwhile, but it prepares us for when cinder will be system-scope aware.\n\nThis commit also bumps the minimum version of oslo.policy to 3.6.2,\nwhich is safer when running tests in parallel with different\npolicy configurations.\n\nChange-Id: I680cb8c4be13bcd3ac6785a7afa81ce5d3477f91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7aea7628b3def5efc8ef2b08d84257c099e0c55b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7aea7628b3def5efc8ef2b08d84257c099e0c55b"}]},"branch":"refs/heads/master"},"118da903b1ff15805ae2efaf397d139145e1cf0c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-17 17:48:52.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/31/775931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775931/3"}}},"commit":{"parents":[{"commit":"9fc6fab79d10a66a8e388d22e7f0ccb736b22eea","subject":"Update secure RBAC check strings with descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fc6fab79d10a66a8e388d22e7f0ccb736b22eea"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-16 19:44:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-17 17:44:40.000000000","tz":0},"subject":"Properly handle InvalidScope exceptions","message":"Properly handle InvalidScope exceptions\n\nNow that we\u0027re starting to set scope_types on default policies, we\nshould make sure we handle InvalidScope exceptions from oslo.policy in\nthe event enforce_scope\u003dTrue. Operators won\u0027t use this switch for a\nwhile, but it prepares us for when cinder will be system-scope aware.\n\nThis commit also bumps the minimum version of oslo.policy to 3.6.2,\nwhich is safer when running tests in parallel with different\npolicy configurations.\n\nChange-Id: I680cb8c4be13bcd3ac6785a7afa81ce5d3477f91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/118da903b1ff15805ae2efaf397d139145e1cf0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/118da903b1ff15805ae2efaf397d139145e1cf0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
