)]}'
{"id":"openstack%2Fcinder~775942","triplet_id":"openstack%2Fcinder~master~I5380d2c96eff8701d4a5702c5cffc92a7d989d55","project":"openstack/cinder","branch":"master","topic":"fc_host_personality","hashtags":[],"change_id":"I5380d2c96eff8701d4a5702c5cffc92a7d989d55","subject":"[Pure] Add support for host personality on FC hosts","status":"ABANDONED","created":"2021-02-16 22:28:15.000000000","updated":"2021-02-17 06:27:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"340fc0a32579c7c7e1a8ddaa5bc9e0edd8e230da","_number":775942,"virtual_id_number":775942,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 22:28:33.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 06:27:47.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"cc47f5a201c9400bc3dc11e63fd0497151164fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-02-16 22:28:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6539a34c112355cba4b5abbb141461ec54c3da5","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-16 22:28:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a95f6624127dd0c9c122814816e5c23309d9b093","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-02-16 22:29:46.000000000","message":"Abandoned\n\nDuplicate","accounts_in_message":[],"_revision_number":1},{"id":"340fc0a32579c7c7e1a8ddaa5bc9e0edd8e230da","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-17 06:27:47.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-775942-1/ns5_nfs : FAILURE in 6h 40m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-775942-1/ns5_iscsi : FAILURE in 3h 49m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bca41844064910c2368b7f823c46ec6f59bd888","revisions":{"2bca41844064910c2368b7f823c46ec6f59bd888":{"kind":"REWORK","_number":1,"created":"2021-02-16 22:28:15.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/42/775942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/775942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/775942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/775942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/775942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/775942/1"}}},"commit":{"parents":[{"commit":"a44aa55fdb5356732e0fdb7e4d527a2853ea3d0d","subject":"Merge \"HPE: Add Peer Persistence support for Primera backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a44aa55fdb5356732e0fdb7e4d527a2853ea3d0d"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-02-16 22:27:20.000000000","tz":-300},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-02-16 22:27:20.000000000","tz":-300},"subject":"[Pure] Add support for host personality on FC hosts","message":"[Pure] Add support for host personality on FC hosts\n\nChange-Id: I5380d2c96eff8701d4a5702c5cffc92a7d989d55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bca41844064910c2368b7f823c46ec6f59bd888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bca41844064910c2368b7f823c46ec6f59bd888"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
