)]}'
{"id":"openstack%2Fcinder~775959","triplet_id":"openstack%2Fcinder~master~Ib7856390053d75bd845476f25891923873bf3078","project":"openstack/cinder","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ib7856390053d75bd845476f25891923873bf3078","subject":"Simplify composite check strings for project personas","status":"MERGED","created":"2021-02-17 02:36:10.000000000","updated":"2021-03-08 01:05:02.000000000","submitted":"2021-03-07 18:05:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775959-secure-rbac","meta_rev_id":"da24b81b0552e9f9df3921d1ef0898dafa8ce3b8","_number":775959,"virtual_id_number":775959,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-07 18:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-02 22:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-02-24 08:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-19 15:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2021-02-19 14:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-07 16:44:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 02:36:42.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 02:51:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 03:23:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 03:34:48.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-17 03:39:43.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-02-17 05:22:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 06:10:53.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-02-17 08:29:31.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 08:45:36.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-17 10:05:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 10:11:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 13:51:50.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-17 17:03:01.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 00:19:19.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-02-18 00:26:24.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-18 17:41:41.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-18 18:09:52.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 19:21:14.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 21:20:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-18 23:28:55.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-19 13:58:23.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-02-19 14:11:42.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-02-19 15:29:03.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-02-20 17:49:47.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-22 06:59:53.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-24 08:40:17.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-03-02 22:43:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-07 15:57:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-07 18:35:59.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-07 22:10:08.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a345ca2d0e6e05ec25a21e7c2f52512d5912f718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-17 02:36:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e59420d66057edfcad193f376da544658309f4ff","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-17 02:36:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46c4ebc5c11aa5ed11bba9215cc0d7095d1b5ad7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 02:51:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/59/775959/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d7e791b/d7e791b.html : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f0ff9c7a009eb7ad2aac1b461735ea9546f6837f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-17 03:23:03.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/1/check/tempest-dsvm-full-focal-py3/c00aff7/ : FAILURE in 37m 10s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/1/check/devstack-plugin-ceph-dsvm-focal-py3/7f3bd27/ : FAILURE in 40m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3f4f0429f7106feacf4f9ece7ca24990a4b4187e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-17 03:34:48.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/59/775959/1/check/dsvm-tempest-TOYOU-ACS5000-driver/8cfc9bc : FAILURE in 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5a1cc42f378e15382557a55a09161561f35ea57b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 03:38:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/59/775959/1/check/iscsi-driver-primera/efc2cd8/efc2cd8.html : FAILURE in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"50091d483d39f74ceb947b2cd408bda12728b842","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 03:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775959/1/iscsi/ : SUCCESS in 1h 01m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/775959/1/smb/ : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5bc4ff9f93b014ce6671547685b9554acae966f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 05:22:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/759d331118414b91a7d3852a0962ac20 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd66a28f73a427ea0261bedd381f4e1 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/927f93de4bc34d9b8ae4e83c3360cb35 : SUCCESS in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ebf66902e264db1b159c52b5de8b817 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b718a6072724a559193904e9319d60e : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383a26f59d744420ad695cafb206dc9c : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/0e30788a8f78452aadc2008bc8c93cea : SUCCESS in 1h 17m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1225348251094f92b59e7aefec35d9f9 : SUCCESS in 56m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/42edb64bec074519a01d5373631640f4 : SUCCESS in 15m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f16d41e02d924b48b5f38dc5583ac6ce : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef3e5230e1204e4d9a225f9d1b8cfff6 : SUCCESS in 8m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7b46930586e6410fb6473365a8f905d2 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9dc2727d8b6c4ec99421031fd51e65a6 : SUCCESS in 1h 01m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca42ae6b36b3414186fe0b57b35b0502 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf283ed50d524b119fac26df85c6480d : SUCCESS in 1h 19m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2071ff65c65346379ba00476f03e909a : FAILURE in 1h 13m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7b14b987c2b943278d6b701f7fc27016 : SUCCESS in 2h 00m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/913f997f00284a7d8df0ae3d39b6474b : SUCCESS in 1h 53m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c19669a1a76f4849817e3179da21586e : SUCCESS in 1h 14m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ad46f216b3024bbf8c9aa0b860d3509f : SUCCESS in 1h 15m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c44b60464d10487398760b0b9ad705d3 : SUCCESS in 2h 40m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7749843d63314fd8a7ed8fa3e264f33d : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6a4bf48f68956f0d9b295f878399f102743f9eb4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-17 06:10:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/59/775959/1/check/dsvm-tempest-cisco-zonemanager-job/81b0ea7/ : SUCCESS in 1h 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"21fd3d20b167b5c98f0edb43d3eb01a7d3d81b27","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-17 08:29:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/59/775959/1/ : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8dcbb874294dd401a43811216f10adcb068694e4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-17 08:45:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/59/775959/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f1e2d2f/ : SUCCESS in 2h 05m 00s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"cc685e02fcefce9e10381ff052fcfac51b3b77ad","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-17 10:05:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/59/775959/1/upstream-check/cinder-cDOT-iSCSI/fff67e8/ : SUCCESS in 37m 04s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/59/775959/1/upstream-check/cinder-cDOT-iSCSI-IPv6/81d2add/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/59/775959/1/upstream-check/cinder-cDOT-NFS/a7f0e02/ : SUCCESS in 37m 11s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/59/775959/1/upstream-check/cinder-cDOT-NFS-IPv6/5f8ebd7/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/59/775959/1/upstream-check/cinder-cDOT-FCP/321cc0f/ : FAILURE in 3h 49m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b98f98af9c774f24303e2a8b04ba506eab6598b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-17 10:11:36.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-1/ns5_nfs : FAILURE in 6h 34m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-1/ns5_iscsi : FAILURE in 6h 28m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"79add1cedc612493b242ee8dc4ef96359a5b89f8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-17 13:51:50.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/59/775959/1/check/DellEMC_PowerStore_iSCSI/c4bf5f3/DellEMC_PowerStore_iSCSI/425 : SUCCESS in 1h 02m 16s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/59/775959/1/check/DellEMC_PowerStore_FC/eb71682/DellEMC_PowerStore_FC/419 : SUCCESS in 49m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"0897dd74fdb5fc14b75108880831f6f0b3c823c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-17 17:03:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"f228b5aed2a60227edf9a155fbf26e5506fd58d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-17 17:48:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"18ba59ff040518f503e1115e9fe0266f5bde9f7b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 19:00:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/59/775959/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a49e577/a49e577.html : FAILURE in 15m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2e1747fb3bcc0297d62560caedfc589654cc5a06","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 19:35:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/59/775959/2/check/iscsi-driver-primera/30d1314/30d1314.html : FAILURE in 1h 01m 04s","accounts_in_message":[],"_revision_number":2},{"id":"dae7fab887053f5ff557482a29b21b8baee48a39","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 19:51:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775959/2/iscsi/ : SUCCESS in 1h 23m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/775959/2/smb/ : FAILURE in 1h 35m 12s","accounts_in_message":[],"_revision_number":2},{"id":"601a645bb0dd45b228da10ba7291bbfec7c3b8a3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-17 19:52:02.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/59/775959/2/check/dsvm-tempest-TOYOU-ACS5000-driver/6f055bc : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5146ada33b5988e7cb0d65bc1b45d7ca3db52af9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-17 21:03:42.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/2/check/tempest-dsvm-full-focal-py3/6edf5eb/ : SUCCESS in 1h 51m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/2/check/devstack-plugin-ceph-dsvm-focal-py3/28ebe93/ : SUCCESS in 2h 10m 14s","accounts_in_message":[],"_revision_number":2},{"id":"53a051a7d38459debaf7c756ecbb3fc31677e446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:23:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32805fa90773459682df93f0c3664b1c : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0167ad09c8c40d6aaee54cb48c70397 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbd67e57dde4471281d2fbb4fd23c1ee : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2964da8e443844a79085c4fc864e5e0b : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e8ced9c4644247ae7df8f18cef297f : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd701106c536473f9e79fee32b5f2155 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/bf2cd056e3ae439399d2e78362d09d38 : SUCCESS in 1h 00m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1e7dce3cbfb41d69ac5fdca1499ad54 : FAILURE in 1h 13m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/842e18fcf0c640fd89234e2c37c97ea7 : SUCCESS in 14m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e3fecf3aa7b4f50bef9c42a0d8998af : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0acf13ce526548988be87d3256eb23cb : SUCCESS in 7m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c8701d9839f4ed5b3d989a95d1627e5 : SUCCESS in 10m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a6db68ae61d3495aba247098452b8f6a : SUCCESS in 41m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/226b479b28a04bc2b3af8989a08e55f3 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c905ac5e34954e1093b9dfd76dc3e24e : SUCCESS in 1h 17m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48055e227bc84818bf96cba870e860ed : FAILURE in 1h 24m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/55aa51bfd5e046a89a4e3372feb5427c : SUCCESS in 1h 36m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5407f522541e40dfa776b94e6863ab83 : SUCCESS in 1h 10m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d095e238fe64e52a0adf7dd7c98cf63 : SUCCESS in 1h 17m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4693375d2baa45e99d0109c9b35e6619 : SUCCESS in 1h 25m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06774866f9d74286b9bf1ce9efe71bcb : SUCCESS in 2h 39m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54e924a53d094adfa48ddfb9a33c18fe : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9df1360ba08a64936bde96582d648d637f166925","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-17 21:25:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"af5926b1ffa9b0871b9c741fe62ef6098128758c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-17 22:23:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/59/775959/2/check/dsvm-tempest-cisco-zonemanager-job/f898b30/ : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":2},{"id":"8ea0dc4bf3939a16ba508090603b1bc780ce2a3c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 22:26:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775959/2/iscsi/ : SUCCESS in 1h 00m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/775959/2/smb/ : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f983157162daaae430c0e23107edf6c5c581adf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 23:33:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04f256d2ca6e458e883e1234655daa44 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338cab5ed35b461db903f23766e3ef82 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e65f9bbcd8d84a20a4aec89b7baef3a2 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eacf11844604ca1afec209c10c5e116 : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9faaedc4a9c47f68cc64b9799b6641a : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3b6de26dcd476fb04c34f9eeb2e84b : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/59e635a2df374b3b807ff5f7d5ad2c7c : SUCCESS in 53m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e20052625d7f4598870b98c3a1115fcf : SUCCESS in 1h 13m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8945caf3f22f41ccb5d62f6e690abb36 : SUCCESS in 13m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cbb1d3b29375479b86f4e60bc61c4089 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a59a29b6372f467eb926c8f00da2c671 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/09f170865c73423ea21b9f1d9a998f60 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/61490877a4b140da82ead7189f702b7e : SUCCESS in 45m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd58e850f8a141cd94f697b2745461d9 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dcb8689f197346f09937180bbff6cd69 : SUCCESS in 1h 36m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e229ae7d21eb497992a3c6ca53858228 : FAILURE in 1h 33m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/baf3ac8471e54ae3b9e5cca5d7344646 : SUCCESS in 1h 29m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fb335af899564a47bbadab8bd563f571 : SUCCESS in 1h 15m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0cc42c647e041babbe8f4175a017eb0 : SUCCESS in 1h 30m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a0babbea6439476085094a53cfa020ae : SUCCESS in 1h 29m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/838d7d8133284d6783960a110f151d8a : SUCCESS in 2h 05m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0b73c9c939f4a38a1c87946ac863f62 : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":2},{"id":"dfa0e7390e5849eff61a6b9bf023222389e665cf","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-18 00:19:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-59-775959-2/1613604813 : SUCCESS in 32m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"2d77298a601c5093965be5dd3a08ddf717dd5677","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-18 00:26:24.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775959-1/PVME-iSCSI : SUCCESS in 48:10.41\n* PVME-FC http://os-logs.tristero.net/dell-ci/775959-1/PVME-FC : SUCCESS in 47:15.17","accounts_in_message":[],"_revision_number":1},{"id":"99f4926dc960a656548426cd04fc812a8bc2db6e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-18 01:23:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/59/775959/2/upstream-solidfire-check/cinder-solidfire-iSCSI/fda554c/ : SUCCESS in 1h 45m 01s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"5764158dc48ba3bdd1a000ac8cb3192dda834677","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-18 01:34:14.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-2/ns5_nfs : SUCCESS in 5h 14m 2s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-2/ns5_iscsi : SUCCESS in 5h 2m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"13abb9d0725afb819254a5af3d3adf7869afc67a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-18 01:48:43.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/2/check/tempest-dsvm-full-focal-py3/144f759/ : SUCCESS in 2h 03m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e4218e51af134979aa782caf9ddcced8 : RETRY_LIMIT in 22s","accounts_in_message":[],"_revision_number":2},{"id":"45c0a7e509e620fa7685bc29fc16fe5793a9666e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-18 02:42:55.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerStore_iSCSI/88795e3/DellEMC_PowerStore_iSCSI/436 : SUCCESS in 52m 45s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerStore_FC/10c09a9/DellEMC_PowerStore_FC/430 : SUCCESS in 48m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"09abf94b45cdab0b6e004eed570b1604e3075ef3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-18 03:47:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-iSCSI/38caec2/ : SUCCESS in 39m 27s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-iSCSI-IPv6/a660bf3/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-NFS/4c73680/ : FAILURE in 3h 14m 46s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-NFS-IPv6/e042666/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-FCP/d31a07e/ : SUCCESS in 1h 17m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cc5c65ae75b2c7a1992841a79d77397908621bf3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-18 17:41:41.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/775959-1/Lenovo-iSCSI : SUCCESS in 46:08.37\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/775959-1/Lenovo-FC : SUCCESS in 42:52.91","accounts_in_message":[],"_revision_number":1},{"id":"e93df34d3c80729aa2c92c927ecc744ae67e4643","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 18:09:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_VNX_iSCSI/f4ab023/DellEMC_VNX_iSCSI/406 : FAILURE in 1h 18m 26s\n- DellEMC_VNX_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_VNX_FC/6c2f05c/DellEMC_VNX_FC/425 : FAILURE in 54m 09s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"359bd581d5aabb3ee0cd309baf1199ba46d6b060","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 19:21:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_XtremIO_FC/3197da4/DellEMC_XtremIO_FC/791 : SUCCESS in 2h 04m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_XtremIO_iSCSI/190f957/DellEMC_XtremIO_iSCSI/793 : SUCCESS in 1h 41m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"68bd6faf3f183af7936935d5b34e49ddb2ea07fb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-18 21:20:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_Unity_FC/79d535c/DellEMC_Unity_FC/729 : FAILURE in 1h 56m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_Unity_iSCSI/ef7e294/DellEMC_Unity_iSCSI/691 : FAILURE in 1h 59m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e88b2b38ee82cb068a4f55b15f16009c882ad6b6","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-18 23:28:55.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775959-1/HPMSA-iSCSI : FAILURE in 1:04:40\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775959-1/HPMSA-FC : SUCCESS in 37:46.98","accounts_in_message":[],"_revision_number":1},{"id":"d42c1aa406d8d1831c1f351051b5379c684955cb","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-19 00:24:12.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775959-2/PVME-iSCSI : SUCCESS in 48:22.71\n* PVME-FC http://os-logs.tristero.net/dell-ci/775959-2/PVME-FC : SUCCESS in 47:29.58","accounts_in_message":[],"_revision_number":2},{"id":"078cd9363a987db0e5edb2988121e6f34d399ee5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-19 05:22:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/59/775959/2/ : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4a2422f86750b069dbd11ef1bb720adb1f68baf3","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-19 13:58:23.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/775959-2/Seagate-iSCSI : SUCCESS in 47:00.96\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/775959-2/Seagate-FC : SUCCESS in 45:17.30","accounts_in_message":[],"_revision_number":2},{"id":"e596dcf57b31bb8f3d146f66e8d729def0a38660","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-19 14:11:42.000000000","message":"Patch Set 2: Code-Review+2\n\nNot a policy expert but this looks ok to me.","accounts_in_message":[],"_revision_number":2},{"id":"8f367fbb86d9e52e8df63b53fe08bd1dc5fab6ac","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-19 15:29:03.000000000","message":"Patch Set 2: Code-Review+1\n\nAs far as I understood this looks good to me but let\u0027s see what other reviews thinks about it 😊","accounts_in_message":[],"_revision_number":2},{"id":"2456648e713e4a80a447042d0f5ea7167b831c5a","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-19 16:58:47.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/775959-2/Lenovo-iSCSI : SUCCESS in 47:56.55\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/775959-2/Lenovo-FC : SUCCESS in 45:03.65","accounts_in_message":[],"_revision_number":2},{"id":"64d7e1b54dbdd3f9f1fc584db5dadd7e2596422e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-19 18:41:36.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e As far as I understood this looks good to me but let\u0027s see what other reviews thinks about it 😊\n\nI think this is going to depend on which personas cinder wants to target for Wallaby.\n\nThe mailing list thread summarizes some approaches we have [0].\n\nThis part is useful if cinder only wants to pursue project-specific personas for Wallaby. Otherwise, if we want to pursue a couple of system personas this release assuming a couple of workaround [1], then these changes probably aren\u0027t needed.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020544.html\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020605.html","accounts_in_message":[],"_revision_number":2},{"id":"e1de43a4958382d92fc111fbd54c7d45fdec3b86","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-20 02:27:26.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775959-2/HPMSA-iSCSI : FAILURE in 1:05:19\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775959-2/HPMSA-FC : SUCCESS in 37:39.76","accounts_in_message":[],"_revision_number":2},{"id":"9769d0f2f63c7e55e36037e50092e8366e425df4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-20 12:36:00.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-2/ns5_nfs : SUCCESS in 6h 54m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-2/ns5_iscsi : FAILURE in 58m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d08c152aed47b4086e8858cfe0d5bc3878bf867c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 17:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/59/775959/2/20210220223533/NEC-FC : SUCCESS in 2h 26m 07s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/59/775959/2/20210220223533/NEC-ISCSI : SUCCESS in 4h 14m 12s","accounts_in_message":[],"_revision_number":2},{"id":"77115dd95e5284ab4fcef2d154181725a6bf55c3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-22 02:58:22.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-2/ns5_nfs : SUCCESS in 3h 54m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-59-775959-2/ns5_iscsi : SUCCESS in 3h 54m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9394ab1cc05e69bded8c2ae5d8d9813a06d27466","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-22 05:23:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"c0fb03ff85a9f62219a84c320a7355552b622205","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-22 06:59:53.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerMAX_FC/20412f8/DellEMC_PowerMAX_FC/486 : SUCCESS in 4h 49m 30s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerMAX_iSCSI/a1c33a4/DellEMC_PowerMAX_iSCSI/450 : SUCCESS in 5h 30m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"17ffacaa22012abe994715f13f674fe5503486c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-22 08:17:11.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"87b8fbc88e934550ff50c1e4ba6a1c2049b1561d","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-24 08:40:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5343cbbee854e46bb7993c0d9f6c95d9ca1eda9c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 22:43:07.000000000","message":"Patch Set 2: Code-Review+2\n\nIt\u0027s worth adding these check strings.  They are clearly named and will make things easier for people configuring policies.","accounts_in_message":[],"_revision_number":2},{"id":"6fab8f3c03b4d76914f5bc153e276a97faf34b8a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-07 14:15:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2b4eb4c02909cbb938fab5dd8bdbcd30427665b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-07 15:36:58.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2fd88bbc97009d7954d32fb75ced761330665717","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-07 15:41:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/59/775959/2/check/dsvm-tempest-cisco-zonemanager-job/42d6fb7/ : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a2e5aab8631b328a0a09d19be751bca43773417c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-07 15:57:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/59/775959/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/acad5f8/ : POST_FAILURE in 1h 40m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/59/775959/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3302871/ : SUCCESS in 1h 36m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d1b4aa67af13a0d4c35aa204b347a45c208a4d8e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-07 16:27:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/59/775959/2/check/dsvm-tempest-zadara-driver/e79ccdd : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 2h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d9786a767e6553ebd09717e1124593cc1a5b820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 16:31:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b168f7378bd545228bc62e1481305dcb : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eeab00b5cdb42d9a9e6de8407296865 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bac023c5768c486db5a39a05e33c9edf : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8656d01a967d490d899377d9498b6ccb : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eb07724468d4ce49844f56de39b8c61 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ba1c4f834c4b72bc34a679db38b384 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/46bdf778a4424e928877f0b1f8f0c04f : SUCCESS in 56m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a07c2b51a4154b9bb077f9056e596764 : SUCCESS in 1h 47m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc309917d14e44d89f18c8cc81ba03b9 : SUCCESS in 13m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f34b95b6959e43c1914e696eb3e105ec : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e4abc0626bc4da1a58974b96937c00b : SUCCESS in 6m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3dc87177db8f45a4ab7297ee932c5eef : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2edfd2058e043468894c5f11b3738d8 : SUCCESS in 46m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75dfd5648ebe4e468c27b2b34b6fea20 : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9cffede8b3814d19bc7b08bd51b7939a : SUCCESS in 1h 12m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8bb097e10e884a9bbaca9b15a568d0b4 : FAILURE in 1h 44m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4cf40c3a0caf481b83bb07f54e56054c : SUCCESS in 1h 27m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e905e322229b46b1bd0eaca5ad223445 : SUCCESS in 1h 12m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1356319b8f954915aa197a7e42332237 : SUCCESS in 1h 36m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e5ca6fa48a5d47098423ee5f54650013 : SUCCESS in 1h 24m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15e718edd13d4304970ec41b3247995d : SUCCESS in 2h 11m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25b56907703543dd97ba5d2316957caa : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"059974cba0137cb09251d6f3a2aa44b35518752d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-07 16:44:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1484689c31621c46e3cc062294a7ca45c38432dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 16:44:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"895be8cc4699fa92307e6fd7c8a7d2891e53ae97","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-07 16:45:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/2/check/tempest-dsvm-full-focal-py3/1b64455/ : SUCCESS in 1h 40m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/59/775959/2/check/devstack-plugin-ceph-dsvm-focal-py3/8be7b16/ : SUCCESS in 2h 03m 08s","accounts_in_message":[],"_revision_number":2},{"id":"074cad5b2f7c94e50db761b9d373f12a0f8c09f4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-07 17:24:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerStore_iSCSI/f44bc97/DellEMC_PowerStore_iSCSI/532 : SUCCESS in 52m 38s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerStore_FC/d7b0d04/DellEMC_PowerStore_FC/526 : SUCCESS in 49m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"4a3fb08d2d2210859e20d16e2fb2caf04233d2ab","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-07 17:36:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-iSCSI/fd6ea2b/ : SUCCESS in 40m 19s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-iSCSI-IPv6/4ed7108/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-NFS/49dabc1/ : SUCCESS in 39m 57s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-NFS-IPv6/93abc5d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/59/775959/2/upstream-check/cinder-cDOT-FCP/bc15482/ : SUCCESS in 1h 03m 37s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a5640bdd78c0a8eef37e361d9628dc9f0b2b8656","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-07 17:42:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_VNX_iSCSI/ad92a42/DellEMC_VNX_iSCSI/505 : FAILURE in 1h 10m 12s\n- DellEMC_VNX_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_VNX_FC/d168205/DellEMC_VNX_FC/528 : FAILURE in 50m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ccf88a799ef700b787c99b80db55cb4e519f155f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-07 17:52:41.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4eba00b5781f4e6d5768fcab80297a8e1f4f988b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-07 18:01:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/59/775959/2/upstream-solidfire-check/cinder-solidfire-iSCSI/5312e96/ : FAILURE in 1h 28m 03s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"df99f3c6bd562cf17a73c03aa1f179788d6cb60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 18:05:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd9183b4b784332a25e49eafc87f599 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c9ffcb9cb824b9ba201cfe67abac6ed : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ce01ad6aca5425cbf0b900e2bb6c031 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2f34e0eb8a54854872a4b5c838cb482 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5cdd063fce4122a972ad7ce60880ac : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/d651aaf27ca4470389b5a2c369761102 : SUCCESS in 53m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb988368fe3f4cd282340b046350df28 : SUCCESS in 1h 09m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/47c3015831b949f796576c6766036161 : SUCCESS in 50m 42s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f838e987739d48e5a9126c2eaae8c504 : SUCCESS in 1h 19m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b73c5dbfa14645975c4ba52ac0d3c4 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c4640dde690d95228ba917dd9e2db0cb88cc0f56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 18:05:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"aa6a9103a87f8bf2f62b953e3fe4b37b908c0726","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 18:06:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0701ec96b0420194061a962e9bb2ef : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2},{"id":"b1d2c3fcddacaaf42ec59752bf47c6a9797cedf7","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-07 18:35:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/59/775959/2/check/DellEMC_SC_FC/e10b622/DellEMC_SC_FC/415 : SUCCESS in 1h 54m 23s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/59/775959/2/check/DellEMC_SC_iSCSI/e7693e4/DellEMC_SC_iSCSI/412 : FAILURE in 2h 03m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"3ed2d0e6105ef3b40e5f838053a78d13f766e2d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-07 20:30:31.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"9b6ec46988e8b958d7e4f700636c7cd138342f14","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-07 22:10:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/59/775959/2/check/DellEMC_PowerFlex/9f938f6/DellEMC_PowerFlex/500 : SUCCESS in 2h 39m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"da24b81b0552e9f9df3921d1ef0898dafa8ce3b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-08 01:05:02.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/775959/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f8e569864f46764fe4befbeb297254082c09bdea","revisions":{"a26bc09e3fc1c74c95b0bad1d3a525011e6bc5a9":{"kind":"REWORK","_number":1,"created":"2021-02-17 02:36:10.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/59/775959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/59/775959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/59/775959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/59/775959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/59/775959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/59/775959/1"}}},"commit":{"parents":[{"commit":"77fed89751fb989561247b8313def557dab972a7","subject":"Make sure we pass context objects directly to policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77fed89751fb989561247b8313def557dab972a7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-17 02:29:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-17 02:29:29.000000000","tz":0},"subject":"Simplify composite check strings for project personas","message":"Simplify composite check strings for project personas\n\nThis commit adds two check strings to cinder\u0027s base policies that\nsimplify project personas.\n\nThe is_admin_project check isn\u0027t used in the admin_or_owner rule. The\nis_admin property of the context object actually evaluates to\ncontext_is_admin, which just checks if \u0027admin\u0027 is in context.roles.\n\nThis check string simplies the indirection. Future patches will likely\nchange many of the admin_or_owner checks to support system-scope, which\nwill require modifications to the existing deprecated defaults.\n\nThe other thing this change does is include project-reader and\nproject-member personas into cinder\u0027s default check strings. While\ncinder doesn\u0027t fully support system-scope, we can still pursue\nproject-personas, allowing a more consistent experience across OpenStack\nservices.\n\nChange-Id: Ib7856390053d75bd845476f25891923873bf3078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a26bc09e3fc1c74c95b0bad1d3a525011e6bc5a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a26bc09e3fc1c74c95b0bad1d3a525011e6bc5a9"}]},"branch":"refs/heads/master"},"f8e569864f46764fe4befbeb297254082c09bdea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-17 17:48:52.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/59/775959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/59/775959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/59/775959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/59/775959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/59/775959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/59/775959/2"}}},"commit":{"parents":[{"commit":"0c87d14f558d78bf0bd436bd74e9e126b20dbe14","subject":"Make sure we pass context objects directly to policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c87d14f558d78bf0bd436bd74e9e126b20dbe14"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-17 02:29:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-17 17:44:40.000000000","tz":0},"subject":"Simplify composite check strings for project personas","message":"Simplify composite check strings for project personas\n\nThis commit adds two check strings to cinder\u0027s base policies that\nsimplify project personas.\n\nThe is_admin_project check isn\u0027t used in the admin_or_owner rule. The\nis_admin property of the context object actually evaluates to\ncontext_is_admin, which just checks if \u0027admin\u0027 is in context.roles.\n\nThis check string simplies the indirection. Future patches will likely\nchange many of the admin_or_owner checks to support system-scope, which\nwill require modifications to the existing deprecated defaults.\n\nThe other thing this change does is include project-reader and\nproject-member personas into cinder\u0027s default check strings. While\ncinder doesn\u0027t fully support system-scope, we can still pursue\nproject-personas, allowing a more consistent experience across OpenStack\nservices.\n\nChange-Id: Ib7856390053d75bd845476f25891923873bf3078\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e569864f46764fe4befbeb297254082c09bdea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e569864f46764fe4befbeb297254082c09bdea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
