)]}'
{"id":"openstack%2Fcinder~779436","triplet_id":"openstack%2Fcinder~master~Iff02775822056904e5457ecddcefb2e580f47953","project":"openstack/cinder","branch":"master","topic":"1918229","hashtags":[],"change_id":"Iff02775822056904e5457ecddcefb2e580f47953","subject":"Bug fix for free space calculation","status":"MERGED","created":"2021-03-09 08:07:06.000000000","updated":"2021-05-03 02:15:45.000000000","submitted":"2021-05-01 07:54:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"779436-1918229","meta_rev_id":"c505325bc110bafc5eb36511478c21380561e8e9","_number":779436,"virtual_id_number":779436,"owner":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-01 07:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-04-12 11:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-28 13:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":1,"date":"2021-04-28 13:54:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-28 13:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 08:13:35.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-09 08:41:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 08:44:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 09:29:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 09:32:02.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-09 09:37:45.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 09:54:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-09 10:10:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 11:15:59.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-09 11:19:26.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 11:25:16.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 11:37:35.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-09 11:59:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-09 12:11:42.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-09 12:13:59.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-09 12:30:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 12:33:15.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-09 19:16:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 22:45:38.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-10 00:03:02.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-10 01:16:57.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-10 17:34:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-25 13:14:29.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 13:40:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-04-09 16:30:58.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-12 06:04:26.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 13:51:14.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-04-28 13:54:19.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"state":"REVIEWER"},{"updated":"2021-04-30 19:30:04.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"2f9a18c977242be86e572041eef34b4b5674eb20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-03-09 08:07:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b46dae4ea1f3465428e4d46b0e5586fb2916c15","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-09 08:13:35.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/36/779436/1/check/dsvm-tempest-TOYOU-ACS5000-driver/58a64f2 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a85e2ac5c6d6f7f9f77b5501ca37e633161b4a54","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-09 08:41:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/763bba818bba419d99311e2d20ec468b : POST_FAILURE in 29m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/1/check/devstack-plugin-ceph-dsvm-focal-py3/d2cad6e/ : FAILURE in 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f6bfd1b4e91585b12979161cd000282ba3d14bfc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-09 08:44:37.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/779436/779436/1/check-not-vote/ext-cinder-zuul/d858f44 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"548233ca6b452c7f4c26034bff7d12c87dd3dffa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-09 09:29:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/1/check/iscsi-driver-nimble/cad8642/cad8642.html : FAILURE in 1h 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e2f838fb05e4e666ec706a0e27a7b90964d7f59a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-09 09:32:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/1/check/dsvm-tempest-cisco-zonemanager-job/a7cb94b/ : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a960ad8403a774f6dd1817c12a538afd8b8385f2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-09 09:37:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/36/779436/1/ : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"205abd53b696fea26d4b9fd7839fc31f26c89d83","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-09 09:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/36/779436/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/dafd95e/dafd95e.html : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"38727ee72d8bb4e886ce79a77ad94eb9a67bbf38","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-09 09:54:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/779436/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d0099b7/ : SUCCESS in 1h 35m 47s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/779436/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e22bfc7/ : SUCCESS in 1h 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"88b76ff19456836a72ae11c174eab1caf7a822ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:10:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28dd894872564d7694d3098fe6752fed : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37e88d35a094a20a8b166bd37b4af66 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/296266452a6c4227b2e21538426ba09e : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d9546e8552649a38e05f385aec68911 : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4705211a50ab4b0097b406c9d5d09555 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da76d2c0a7b46899558e33fba63c74d : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/6be536ea0785482e95be2f58c6c72f80 : SUCCESS in 1h 05m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abf03d8c2f8149e2972a700534466200 : SUCCESS in 1h 19m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3243203b04db448f9ce6869ded152f9c : SUCCESS in 6m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/679add1fd4c849a0a154fa018a0e94b8 : SUCCESS in 16m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3dc5d673fe67474aa571cc237575a792 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddee8633b2a349c58125e3009cb5d5f0 : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd8e9c74f4754ed09b0121cfefef4fe8 : SUCCESS in 9m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b7ab10a85f5348eb88c6075f0d47fc67 : SUCCESS in 49m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0fd5506a77849c28facbd0cc5b25492 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d11c80ef990241dea24a5a5728ec08a9 : SUCCESS in 1h 30m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3bcea212d77846cdb15b466f5b1f593f : FAILURE in 1h 24m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6273bf4c729e44af8f62b964fe6368a9 : SUCCESS in 1h 38m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/30d6ede2d93d4b069362225c69343e37 : SUCCESS in 1h 16m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/667cbb8f521d460d83944acc27302c96 : SUCCESS in 1h 24m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbe2036efbde4d899acace58d347ae21 : SUCCESS in 1h 30m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17b7a94f99284e0dbb52f6c91e83b3db : SUCCESS in 2h 01m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cadcaaad48d4f1683b2e6acbf6e666b : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b984ac18e9b9211e3cf31936e66a40d976c4a03b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-09 10:16:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/36/779436/1/check/iscsi-driver-primera/e33fba7/e33fba7.html : SUCCESS in 1h 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"093e65773490106369375c58a557b5c462b08a80","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-09 11:15:59.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/36/779436/1/check/DellEMC_PowerStore_iSCSI/522ee0b/DellEMC_PowerStore_iSCSI/546 : SUCCESS in 1h 04m 46s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/36/779436/1/check/DellEMC_PowerStore_FC/09df6c6/DellEMC_PowerStore_FC/540 : SUCCESS in 50m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"be1ec8c909925aee962d686d4981a62860c4b03c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-09 11:19:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/36/779436/1/upstream-check/cinder-cDOT-iSCSI/34a7476/ : FAILURE in 57m 42s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/36/779436/1/upstream-check/cinder-cDOT-iSCSI-IPv6/70394d0/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/36/779436/1/upstream-check/cinder-cDOT-NFS/62b5df6/ : SUCCESS in 38m 39s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/36/779436/1/upstream-check/cinder-cDOT-NFS-IPv6/9138250/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/36/779436/1/upstream-check/cinder-cDOT-FCP/c3a27e3/ : SUCCESS in 1h 07m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7cff3f3b075352595ad1c5c3fc479a74b371921a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 11:25:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/36/779436/1/check/DellEMC_VNX_iSCSI/0fbc319/DellEMC_VNX_iSCSI/519 : FAILURE in 1h 14m 07s\n- DellEMC_VNX_FC http://publiclogs.emc.com/36/779436/1/check/DellEMC_VNX_FC/231cd4f/DellEMC_VNX_FC/542 : FAILURE in 52m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e415d5313c92e6b8b9eac557975aa37d225c408b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-09 11:37:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/36/779436/1/upstream-solidfire-check/cinder-solidfire-iSCSI/2d1193f/ : FAILURE in 1h 25m 33s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"19ec2fb6d94a87cbd5a0ee9691262b98cf8ab950","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-09 11:59:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/1/check/DellEMC_Unity_FC/ca04af6/DellEMC_Unity_FC/839 : SUCCESS in 1h 42m 38s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/1/check/DellEMC_Unity_iSCSI/05d36f5/DellEMC_Unity_iSCSI/794 : SUCCESS in 1h 48m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4fff3112c29d1e4180eab48c71141f9663c19e6e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-09 12:11:42.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/779436-1/Seagate-iSCSI : SUCCESS in 43:08.73\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/779436-1/Seagate-FC : SUCCESS in 41:56.45","accounts_in_message":[],"_revision_number":1},{"id":"de8b3b028d74adf7e906a38281b5f863c18d475c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-09 12:13:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/36/779436/1/check/DellEMC_SC_FC/b0281b5/DellEMC_SC_FC/429 : SUCCESS in 1h 58m 31s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/36/779436/1/check/DellEMC_SC_iSCSI/4a1fd85/DellEMC_SC_iSCSI/426 : FAILURE in 12m 05s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6b0a1aead16fea8be11ce31c069bf75c28fb66e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-09 12:30:22.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-1/ns5_nfs : SUCCESS in 3h 21m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-1/ns5_iscsi : SUCCESS in 3h 17m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"2ff6a96d250e4ebe701f9227fb7d23d7e1983361","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-09 12:33:15.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/779436-1/HPMSA-iSCSI : FAILURE in 1:03:46\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/779436-1/HPMSA-FC : SUCCESS in 41:00.35","accounts_in_message":[],"_revision_number":1},{"id":"8f0a0bb797dace4ba3ff27a40a4f84e8d5b53be7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 19:16:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/36/779436/1/check/DellEMC_PowerFlex/8bbb264/DellEMC_PowerFlex/514 : SUCCESS in 2h 45m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9f2e1c27e3859cb268e4ed6485624efd3be95c30","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-09 22:45:38.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/779436-1/PVME-iSCSI : FAILURE in 3:09:00\n* PVME-FC http://os-logs.tristero.net/dell-ci/779436-1/PVME-FC : SUCCESS in 55:05.29","accounts_in_message":[],"_revision_number":1},{"id":"211846fd3d460a7cc59c7930b564b4656e4217f3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-10 00:03:02.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/779436-1/Lenovo-iSCSI : SUCCESS in 2:16:00\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/779436-1/Lenovo-FC : FAILURE in 4:47:35","accounts_in_message":[],"_revision_number":1},{"id":"562861f41f4b540ad1247fc78b6206419be88041","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-10 01:16:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/1/check/DellEMC_XtremIO_FC/ad1ea89/DellEMC_XtremIO_FC/924 : FAILURE in 2h 02m 28s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/1/check/DellEMC_XtremIO_iSCSI/0031ad0/DellEMC_XtremIO_iSCSI/935 : SUCCESS in 1h 48m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f782f4e16eb147a700e055a7510bc48bc457a9e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-10 17:34:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/1/20210311000123/NEC-ISCSI : SUCCESS in 2h 30m 35s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/1/20210311000123/NEC-FC : SUCCESS in 2h 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cf96bd522522a042a0d43d662e4bc63844b6866b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-03-25 05:53:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f81a0ddf3282ecdda3f0a947849a2ffa986e9501","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-25 05:55:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a215cff0ddd9d477a1b9c62c73b9091b960b91e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 06:21:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfde67b4033d4a598cac7331312a0dcd : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82d8b3697760435f8d9273e7014c5063 : SUCCESS in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d2f8aa3c745c1fa716ae12600b0a39b879ba792","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-25 07:56:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_Unity_FC/0ecb1b7/DellEMC_Unity_FC/969 : SUCCESS in 1h 54m 33s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_Unity_iSCSI/0b978b5/DellEMC_Unity_iSCSI/911 : SUCCESS in 2h 02m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"83fcc59a5bacfeea70b43995d1a708de1f2dc43f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:04:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3737aff40c24470ab255e7a8f87a8ded : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548e5c7f3a4947258ad245ed219b0fc4 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4dbf4df28424b59b577e216d1a802a0 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88478af7e3574c4394770bedf982e139 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/930324a67b9d41d4bf2b9751ca9a67f4 : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6229169b25a455a9731d550efedc120 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/d2fdb188561644a397c69c001cde8b9f : FAILURE in 31m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/624bc09cd9d04f3ea92bd593caec72b0 : SUCCESS in 1h 29m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ea7bcf5d3da40cf87c873a7444b309c : SUCCESS in 7m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2984800713e7497f9cf3dec4b6a6eeac : SUCCESS in 15m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89852eb4b9884a618b5c5d75f7a09804 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e784c65b49d44908acc7cfe64809f285 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3df7e0f94f8a4f2aafeb4e9f5764e562 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0c05024bdec46d2abe6e72d5846d202 : SUCCESS in 45m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c11b74fe0034e09b02a8b9c138128a3 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b821a62b7cff4105a382d0abb2b9afd6 : SUCCESS in 1h 17m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a7f93d25eb04f5b93941baa7af930e5 : FAILURE in 34m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dfd4b3dbd2854466a89a402f32caede0 : SUCCESS in 2h 04m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee5f9bd4127849d39e7196767a4fd054 : SUCCESS in 1h 20m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3f868571c57406a872838fd2e0b02dd : SUCCESS in 1h 31m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc4639ff008545c7aff62bae2fa2a06b : SUCCESS in 1h 20m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d8ad690a15a40baae41ae4f7062e83a : FAILURE in 33m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ada661fca19246ef821a37c0bb4d03a8 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":2},{"id":"2a551f28db4664768db2e1cf7271bf971db559f5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-25 08:42:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/2/check/dsvm-tempest-cisco-zonemanager-job/1a99b5a/ : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a116c993fdf1138dd614ec998551c9bc08fb74de","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 11:02:26.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-2/ns5_nfs : SUCCESS in 5h 8m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-2/ns5_iscsi : SUCCESS in 4h 54m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"856f0df354f9740a3288fa31f2cd81bfa16facd3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-25 13:14:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/36/779436/2/check/dsvm-tempest-zadara-driver/16e4949 : SUCCESS in 1h 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f740467e048a92bbbe4663d76c776d0a6f747d2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-25 14:43:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/2/check/iscsi-driver-nimble/3bd0cdc/3bd0cdc.html : FAILURE in 1h 20m 00s","accounts_in_message":[],"_revision_number":2},{"id":"84a2523c0283e6fbdd35c916eb5be0fab2ab31c7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-25 15:18:03.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_XtremIO_FC/91e5b96/DellEMC_XtremIO_FC/1126 : SUCCESS in 2h 03m 11s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_XtremIO_iSCSI/db4cc44/DellEMC_XtremIO_iSCSI/1144 : SUCCESS in 1h 45m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"49348bafc20bdd43b739ba318a29f857b3a6e47a","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-03-25 17:44:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29db6c1bfd7dcc8a6cb994a98a5ff727232750c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-25 17:45:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4efeaa7e7baffcda37dae31a2d9660c4353836a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f09f8f0a38549d29604c5689ea62e29 : SUCCESS in 20m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0487997c9754657b18ec2ac67775031 : SUCCESS in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e33047d5cef100c30ad92915b89ba6a67b709ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 20:01:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d41139f573944b999ad6cce31085418 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b36c4b6b0fe45a098e8e6fd140ebc82 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffc677afe8e748489b10b6f98a1bb2a2 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dd09af5ed22465b90511acf773cde62 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9c6681aee7a444a83684d5a47a8ff0a : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3197d6f29242f194c77eff133da173 : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/9e83426747724fa5bdd415c8383fc2d7 : SUCCESS in 1h 03m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/beab02331a254b238646a282a2ce08d3 : SUCCESS in 1h 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50379947f5a74ff2b9b825852a10b50d : SUCCESS in 7m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/285e4cc589b5455aacd5b7705092f7f6 : SUCCESS in 13m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d36acf3aad6b4201b89af210ee2f3df5 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bf1a0fad66e486b8471e8ca5abf18b9 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98448347ea79445c9034499f502b5bfe : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/745c22fe8ced459cb718b94222e6a597 : SUCCESS in 51m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2de1e16693814f85b972d7dbb5c85810 : SUCCESS in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec7dc2d6a6844d26bef995e93d5241a1 : SUCCESS in 1h 24m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/37160ba6eaea4dd9aeee0d55d1933590 : FAILURE in 1h 41m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f7c5455888e48299be92297371cb937 : SUCCESS in 1h 29m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fe531d59974e4273baa00e269b8675b0 : SUCCESS in 1h 21m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef1c9512b5de44e883d3dd6f33b2ecce : SUCCESS in 1h 24m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33b554c0efd94da29116dbee262af491 : SUCCESS in 1h 35m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6d0ca4c03ff42559579758ec3c3bfa4 : SUCCESS in 2h 05m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bebf19e46e94863b794dcaa76f8d576 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fbd6f13d3d8855e34ffa2a15775fd0c2367a956b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-25 20:16:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/36/779436/2/upstream-solidfire-check/cinder-solidfire-iSCSI/d2c3fbd/ : FAILURE in 13m 51s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"457bc47fd7dcfdae420211d70e3f43ff8c033c7d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-25 20:17:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/36/779436/2/upstream-check/cinder-cDOT-iSCSI/d92429f/ : FAILURE in 14m 16s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/36/779436/2/upstream-check/cinder-cDOT-NFS/26ca2fa/ : FAILURE in 14m 16s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/36/779436/2/upstream-check/cinder-cDOT-FCP/de04cb0/ : FAILURE in 14m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e0c5986261f15f2d66585b16353a3b844a37666","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-25 20:32:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/2/check/dsvm-tempest-cisco-zonemanager-job/9f3be21/ : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0ecdfb3c498f6d1cbbe59058f76d39ed14d525c3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-25 21:43:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_PowerStore_iSCSI/5ea50c2/DellEMC_PowerStore_iSCSI/713 : FAILURE in 45m 53s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_PowerStore_FC/0a5c5b4/DellEMC_PowerStore_FC/707 : FAILURE in 37m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2d9a68fc185140ce206f82c6a3f89a2a4754339","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-26 03:14:43.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/779436/779436/2/check-not-vote/ext-cinder-zuul/fad076f : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9f5ffff47a21771fa44066d6292b4a4b76373304","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-26 03:44:23.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/2/check/tempest-dsvm-full-focal-py3/7a5e0c8/ : SUCCESS in 1h 33m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/2/check/devstack-plugin-ceph-dsvm-focal-py3/4a15ea9/ : SUCCESS in 1h 49m 01s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud finger://zuul-v3-executor.novalocal/a7c34bdc3bc44804aa9ebfd44bb36dc3 : TIMED_OUT in 2h 35m 20s","accounts_in_message":[],"_revision_number":2},{"id":"67d0a00d46c2f7260ba90c72517391800cdd59d7","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-03-26 03:56:56.000000000","message":"Patch Set 2:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":2},{"id":"5a17795a4b957626618c80eef5dd78b4035224cc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-26 05:29:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/2/check/iscsi-driver-nimble/44b807c/44b807c.html : FAILURE in 47m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ded0c5f87874966534a8e705a2ae320b2619e1c9","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-03-26 08:28:00.000000000","message":"Patch Set 2:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":2},{"id":"6c9f8072466ee5246cc5942fbda3d06385d5ae96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-26 10:31:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/2/check/iscsi-driver-nimble/c9b17a4/c9b17a4.html : SUCCESS in 2h 02m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e40201dda9655ed702483a1740bfa7b57f56a003","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-26 18:03:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_Unity_FC/796fa23/DellEMC_Unity_FC/986 : SUCCESS in 1h 39m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_Unity_iSCSI/5d8d77b/DellEMC_Unity_iSCSI/928 : SUCCESS in 1h 52m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"324e6e828bd92d961cc184d988a62af2304e4618","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-26 18:15:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/36/779436/2/check/dsvm-tempest-zadara-driver/4e6b41b : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 2h 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e65b931c0e4d949c8a5afcc63509689c3ee445d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-26 19:16:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_SC_FC/0150989/DellEMC_SC_FC/512 : FAILURE in 2h 24m 57s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_SC_iSCSI/76f2a74/DellEMC_SC_iSCSI/509 : SUCCESS in 2h 06m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"aef7873c08bb4ea570cbd24034f024888f67bf34","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 22:12:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/36/779436/2/check/DellEMC_PowerFlex/01b784a/DellEMC_PowerFlex/666 : FAILURE in 2h 46m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"2dc1de4f152ddd36ba7a2f6a3c343a51c9fc2f66","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 23:02:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/2/20210327053844/NEC-ISCSI : SUCCESS in 2h 17m 40s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/2/20210327053844/NEC-FC : SUCCESS in 2h 23m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a4436908fe2761d8a40f5c15a6cc805a5159f0b7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-27 01:51:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_XtremIO_FC/3a54847/DellEMC_XtremIO_FC/1142 : SUCCESS in 2h 04m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_XtremIO_iSCSI/4c01d32/DellEMC_XtremIO_iSCSI/1160 : SUCCESS in 1h 45m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"e918bf370322030d9158b899107c2382a2cde591","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-27 11:40:36.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/36/779436/2/check/DellEMC_VNX_iSCSI/e81c0e3/DellEMC_VNX_iSCSI/647 : SUCCESS in 3h 18m 17s\n- DellEMC_VNX_FC http://publiclogs.emc.com/36/779436/2/check/DellEMC_VNX_FC/9bbe426/DellEMC_VNX_FC/693 : SUCCESS in 1h 24m 35s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"11eaf1d535cd0501c8b8a2a6bf471494c310fe2d","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-27 13:19:01.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/779436-2/PVME-iSCSI : FAILURE in 1:41:59\n* PVME-FC http://os-logs.tristero.net/dell-ci/779436-2/PVME-FC : SUCCESS in 56:40.21","accounts_in_message":[],"_revision_number":2},{"id":"597fc2597210aa80e19d77e3ea202274975d81fc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-09 13:40:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nCode change and tests look good.  See comments inline about the commit message and release note.","accounts_in_message":[],"_revision_number":2},{"id":"b8332542938dfd7b42387f5ec8c813504e282510","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-09 13:44:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48eca930d671012ca6f44e08a6291977c1e269e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-09 14:58:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a4f584a025ab07bacd5143744df0499b370b5e3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-09 15:46:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nNit noted inline that you can fix real quickly.","accounts_in_message":[],"_revision_number":3},{"id":"2a2a325a03971b172348db9bf59d5bde84199887","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 15:55:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/36/779436/3/check/iscsi-driver-primera/69e4010/69e4010.html : FAILURE in 55m 00s","accounts_in_message":[],"_revision_number":3},{"id":"4f248dae9b165029f3829a113039d44c27c5df11","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 16:10:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/3/check/iscsi-driver-nimble/2d4b9e3/2d4b9e3.html : FAILURE in 1h 10m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d8074285193bcdcf417a322988cfe3bfa215b526","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-09 16:15:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/3/check/dsvm-tempest-cisco-zonemanager-job/75b5bdc/ : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b3ad556e56a98610c4496e30f0490c49d11e5445","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-09 16:30:58.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/779436/3/iscsi/ : FAILURE in 1h 31m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/779436/3/smb/ : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":3},{"id":"d385d4305542417815324863a5548bdc7106d580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 18:42:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/727cf61fb46f4cf5bc7462034eb9fe7c : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122cb764aeb341f6a82e0042c0065a4e : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb05c7f17c4446859b030e5f0949cf6a : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3767b634e69d4b7181c4e1d9197ccbab : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc44fb53f5344fbd87f86d5e89c7db3c : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1de25f271b40d6a656af4b081474d5 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/e80a0e51860045f5b0ad7bc5cd9a04af : SUCCESS in 56m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32e9068207e44f7cbbc3429d680a3f05 : SUCCESS in 1h 14m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c67d9ae3265c47df892f5b2e7968ace5 : SUCCESS in 7m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3182a4ba8bf84e548ee757583f094a81 : SUCCESS in 15m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3541283bca3406d8899807adce862a3 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32ed10dc72874aa4b845866e210c59e5 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/246fa337c3564ee7929ecf06784fa96f : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f55db098aa54f979addc03b500e9431 : SUCCESS in 30m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13a1c03343e04835b2d7f29ed1cb7978 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb679531692b40c298dabc080c3e6d4c : SUCCESS in 1h 09m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/83d548ecab0441bd9a377a3ae8a49ad8 : FAILURE in 28m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/430bac03c3704f6290d4567c5666d0b7 : FAILURE in 1h 29m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5fd63a7d85c45b188ae7eed3e5b8858 : SUCCESS in 1h 00m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fca936e7074749b99944933d86f19856 : FAILURE in 9m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d8b1f8dfcabe4cb7b0e58b2ce9176898 : SUCCESS in 1h 29m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5dfac0ea9244ce2b2c9a33db3908c66 : SUCCESS in 2h 09m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b274d9532dc24546b8a5de3df1e632d7 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":3},{"id":"605ece8c6818def06f4395ea70762a21bc6e95b9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-09 19:20:56.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/3/check/DellEMC_Unity_FC/5091889/DellEMC_Unity_FC/1106 : SUCCESS in 1h 24m 24s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/3/check/DellEMC_Unity_iSCSI/2b7e4e4/DellEMC_Unity_iSCSI/1043 : SUCCESS in 1h 50m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ef8aeeac5879a91b15f8df53692529e8ce10fbd2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 19:56:27.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/3/check/DellEMC_XtremIO_FC/240de6d/DellEMC_XtremIO_FC/1277 : SUCCESS in 1h 58m 53s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/3/check/DellEMC_XtremIO_iSCSI/ea3d70f/DellEMC_XtremIO_iSCSI/1299 : SUCCESS in 1h 42m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"82c285f09006df85e9ee722b711fe621976454d8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 21:20:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/3/20210410020151/NEC-ISCSI : SUCCESS in 2h 10m 29s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/3/20210410020151/NEC-FC : SUCCESS in 4h 18m 23s","accounts_in_message":[],"_revision_number":3},{"id":"07c092fb12278fefecaad4a787fe163ce9018ca8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-09 21:45:28.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-3/ns5_nfs : SUCCESS in 4h 58m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-3/ns5_iscsi : SUCCESS in 4h 57m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8a175d6db6786565e1d07019e69da330567dfaa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-09 23:15:30.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"70ccb3625e09ff7f5abf101ddcdcad8b290a18da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 23:16:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"a5d1f6715a0427e4ba738eefe89472def1c58f6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 23:34:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/291577e586b2466ba17703629962e7de : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d8aee0c82d14ee3996b02ae12e62218 : SUCCESS in 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00c54b2492a0ed30a6f2f796d366e291df873b37","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-10 00:15:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/779436/4/iscsi/ : SUCCESS in 59m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/779436/4/smb/ : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":4},{"id":"6b1dd36dc431a5a070df3abc86a0484c24ba1387","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-10 00:16:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/36/779436/4/check/iscsi-driver-primera/4ed009f/4ed009f.html : FAILURE in 59m 18s","accounts_in_message":[],"_revision_number":4},{"id":"515493a9858e586f300bb55281d1ed5900c747af","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-10 00:31:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/4/check/dsvm-tempest-cisco-zonemanager-job/ad53778/ : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":4},{"id":"973bcefc783eab0c15ab94e1413b7f9075faf6f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-10 01:06:22.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_Unity_FC/15160de/DellEMC_Unity_FC/1109 : SUCCESS in 1h 28m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_Unity_iSCSI/eecf798/DellEMC_Unity_iSCSI/1046 : SUCCESS in 1h 50m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"42beb7b8c424512c431cbee7510893e470a68e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 01:22:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64f22d2e5ead4fbe96f7824ebff38a1a : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e2249ec13a45298bf95d6768157903 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa7bc6b36f3f4f9182af3223593928d7 : SUCCESS in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cfec6914bff42358cd3108a92d6fcc7 : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20e5b808184a4ddca03a17addd66ed4a : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d341eecec5634d51aaf74cffe083d86c : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/59a0ef5b32e04974af80d865764c0039 : SUCCESS in 50m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/819974d271384da3b4348b068e156e57 : SUCCESS in 1h 07m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26369353066c46cc921f737214b992c5 : SUCCESS in 8m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/984e0593324045088182267b29c09f20 : SUCCESS in 12m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/899e5539d9314d718b38a4fda3265805 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0448f2ed789b4f2f88611171f124393a : SUCCESS in 6m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19fa17e241854f99a8ab8917bf9de954 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9e24d057f946420ca72a95a612bf61dd : SUCCESS in 42m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/658ea8fb01154510a688fcdeae72ef57 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d8684117f1074e9f8335801b95d24b98 : SUCCESS in 1h 18m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/23b8e5acac0d4f15b7341752e6b7cbcc : FAILURE in 34m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef4459810b9a4edfb7335a9235af31d7 : FAILURE in 1h 27m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8922d4d7db7549c9b9e55dee7aebb9b5 : SUCCESS in 1h 05m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d87c53bb3ca24df782fc522e8dd3b9a1 : SUCCESS in 1h 16m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a7828d068fa4fc9871c8182b53956f9 : SUCCESS in 1h 15m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbfa179b44b74daab14ec7da8cc2bdfb : SUCCESS in 2h 01m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dccf80af3e1a4bfea8af2fd8d9062fbc : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":4},{"id":"3c258d8674782ec93da61b066606c2b531b6150e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-10 01:54:55.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_XtremIO_FC/e004b8b/DellEMC_XtremIO_FC/1280 : SUCCESS in 1h 57m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_XtremIO_iSCSI/432a96c/DellEMC_XtremIO_iSCSI/1302 : SUCCESS in 1h 42m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"db6b68b950a84a4af6473059a4558c8ee473c5b9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-10 05:23:04.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-4/ns5_nfs : SUCCESS in 4h 51m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-779436-4/ns5_iscsi : SUCCESS in 4h 41m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"982c677ff3093f9bc3b378cf9ca683c42c1abe75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-10 18:42:38.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/4/check/tempest-dsvm-full-focal-py3/92440fa/ : TIMED_OUT in 3h 08m 51s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/4/check/devstack-plugin-ceph-dsvm-focal-py3/42d6203/ : TIMED_OUT in 2h 36m 33s","accounts_in_message":[],"_revision_number":4},{"id":"238a9fd913c4d4daa9e5ac7ce2ab729c62da5cda","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-12 00:33:02.000000000","message":"Patch Set 4:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":4},{"id":"e1a1548031199ceee388239c6ee8bbd099ee167a","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-12 00:33:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f340b72b6f99643dea50db3214cc5dc8f6cfd344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 00:53:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db19910e69934d38bb1632571eedd4c8 : SUCCESS in 15m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddf1392ad72a45b2bc1920e809a93fbf : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"56ee800ffcf7bff785b2ab020333c2676a51795f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-12 01:47:55.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/4/check/dsvm-tempest-cisco-zonemanager-job/5ef46ba/ : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":4},{"id":"02e0fff1283552b010654dd399e31a2b7e6e23ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 02:13:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e51d04dff088463799e9f3c7bf291aca : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b555982cb54e40f1b8c97c817dca817e : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4bf195bd01e432a92e671af05eb8ec6 : SUCCESS in 10m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/827d9a5edb364da3b1f7e2a702a7b58b : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1ee7f4f2dc4d05b3f65ade3c3b216a : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196485c97430488b8032be30419e0188 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3dafacb7f5ca4212b6e64744de2ad379 : SUCCESS in 35m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29738f122bb945c191f8e1f38d3bbf1d : SUCCESS in 1h 06m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afdfa1e365e848b19f7989989de76031 : SUCCESS in 5m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08eac861a112400aa5a93ff17d2dfa90 : SUCCESS in 13m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c3173120ef54db59c96eb00b3dc5d97 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/604df91a39134361a74ff97bb20738ea : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a865f8b6a914435b4fffa9911bc864d : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3e30afc6b858423096a1c08d267db8d5 : SUCCESS in 33m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca8b60b411f04f78895045eca4c56153 : SUCCESS in 4m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36def3cea3a04ae0bdc53e34953ac566 : SUCCESS in 1h 04m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a66ff84249841a1b8a15037e1c1b5bc : FAILURE in 21m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae69a321bfca4b9db94b56be21c6dde4 : SUCCESS in 1h 26m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1edc1617e6c45ac884a750d8764cb06 : SUCCESS in 43m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25222f27710147658603a48504cadd3e : SUCCESS in 1h 12m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2bd05af125c44ecab936dbc0f5f2f551 : SUCCESS in 57m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/63c651b63b204dcf8d56e5b7190cbfb3 : SUCCESS in 1h 36m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78b4b1599eec44ac95888c526f6b6499 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":4},{"id":"e1da9260049f6d072658f6a4fb2e483550c670de","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-12 02:40:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/4/check/iscsi-driver-nimble/779810c/779810c.html : SUCCESS in 2h 05m 41s","accounts_in_message":[],"_revision_number":4},{"id":"cd234a094d0d02ac615cd714647b86559e618fa6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-12 03:09:41.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerStore_iSCSI/0b392d5/DellEMC_PowerStore_iSCSI/768 : SUCCESS in 56m 00s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerStore_FC/2e7786c/DellEMC_PowerStore_FC/763 : SUCCESS in 44m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"f394a2e1923e1a1b218d3e3bb12489167aa27c07","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-12 03:36:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/36/779436/4/upstream-solidfire-check/cinder-solidfire-iSCSI/469e346/ : FAILURE in 1h 22m 04s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"787b9328d66a0d37bdc62ebe1f7f566686468439","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-12 03:58:43.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_Unity_FC/f0e0cbe/DellEMC_Unity_FC/1111 : SUCCESS in 1h 21m 50s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_Unity_iSCSI/e868a1f/DellEMC_Unity_iSCSI/1048 : SUCCESS in 1h 45m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"959bacb369f06b72c1516f830069255355fa47f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-12 04:10:28.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/4/check/tempest-dsvm-full-focal-py3/45b75a2/ : TIMED_OUT in 3h 09m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/779436/4/check/devstack-plugin-ceph-dsvm-focal-py3/4776138/ : TIMED_OUT in 2h 37m 21s","accounts_in_message":[],"_revision_number":4},{"id":"d49cd0b227c382bf6324e4ebf51f01cca966d1b2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 04:12:11.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_XtremIO_FC/0239f8f/DellEMC_XtremIO_FC/1282 : SUCCESS in 1h 58m 35s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_XtremIO_iSCSI/3ca50e7/DellEMC_XtremIO_iSCSI/1304 : SUCCESS in 1h 43m 52s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"d98bc8e8145235e1b1a0a48a649787eecc73ce8d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-12 04:16:40.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/779436-4/Seagate-iSCSI : SUCCESS in 46:28.74\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/779436-4/Seagate-FC : SUCCESS in 39:50.51","accounts_in_message":[],"_revision_number":4},{"id":"7bf31b8549df1e217b1cc7098f684e6f28e677d9","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-12 04:23:09.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/779436-4/Lenovo-iSCSI : SUCCESS in 48:09.34\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/779436-4/Lenovo-FC : SUCCESS in 45:05.09","accounts_in_message":[],"_revision_number":4},{"id":"fddcc6f46b46ecb26f34ac12f2e461fd1224b4cc","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-12 04:29:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_SC_FC/3228d89/DellEMC_SC_FC/610 : SUCCESS in 2h 03m 31s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_SC_iSCSI/53351ad/DellEMC_SC_iSCSI/607 : SUCCESS in 2h 15m 57s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"100d811cca67ec800cb5537ce1fb69fc7e313b25","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-12 04:32:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/36/779436/4/upstream-check/cinder-cDOT-iSCSI/67406af/ : SUCCESS in 35m 53s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/36/779436/4/upstream-check/cinder-cDOT-NFS/f19a385/ : SUCCESS in 2h 01m 21s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/36/779436/4/upstream-check/cinder-cDOT-FCP/ea8dd33/ : SUCCESS in 59m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3daa39d33eda07bcbc46cff691e4474d79db6c90","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-12 04:37:47.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/779436-4/HPMSA-iSCSI : FAILURE in 1:04:07\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/779436-4/HPMSA-FC : SUCCESS in 33:30.67","accounts_in_message":[],"_revision_number":4},{"id":"e9370111e9f225e881f04d42e9d61d36be51aae1","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-12 04:38:37.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/779436-4/PVME-iSCSI : SUCCESS in 56:08.59\n* PVME-FC http://os-logs.tristero.net/dell-ci/779436-4/PVME-FC : SUCCESS in 55:52.09","accounts_in_message":[],"_revision_number":4},{"id":"14984a0cb8d6272ef9c606ea0b32cbcdcf60a849","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 04:49:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerFlex/b952cdd/DellEMC_PowerFlex/756 : FAILURE in 2h 35m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"8d215d38d476240cdd7957750ee7a7adb77e8b78","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 05:38:36.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_VNX_iSCSI/4b50879/DellEMC_VNX_iSCSI/736 : SUCCESS in 3h 25m 03s\n- DellEMC_VNX_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_VNX_FC/c71ff5f/DellEMC_VNX_FC/785 : SUCCESS in 1h 11m 54s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"f7442195efa39ff4a1ac2cff6bad9bb9cb0dd19b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 06:04:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerMAX_FC/5eca727/DellEMC_PowerMAX_FC/701 : SUCCESS in 5h 04m 37s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerMAX_iSCSI/b571508/DellEMC_PowerMAX_iSCSI/653 : FAILURE in 9h 21m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"db8508260d6a0cb0f18d208edcf938070ed7095e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 09:50:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/4/20210412163406/NEC-FC : SUCCESS in 2h 10m 24s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/779436/4/20210412163406/NEC-ISCSI : SUCCESS in 2h 16m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d5d6e9c0005f2170f1cd5f5bf19dd8047081cd3b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-12 11:53:10.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"68694e7a77795f057ef38e5875278bec6ca42985","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-28 13:51:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"47dbf4336ad692deec6a64dd60108ca2cb4b2eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:51:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e96b872848c438147b20370235e3a17aa5b55cb7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-28 13:52:56.000000000","message":"Patch Set 4:\n\nI noticed the nimble CI is passing[1] but not reporting logs in a readable way. See other driver CI reports\n\n[1] https://hpe-storage.github.io/nimble-ci-logs/logs/36/779436/4/check/iscsi-driver-nimble/779810c/779810c.html","accounts_in_message":[],"_revision_number":4},{"id":"613c35dc2c204085b9d75e84444984a69f1c5a5a","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-04-28 13:54:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1290d2f5a4c4f7e304035f0778795ec813cb7f29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:24:56.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75be3e6343bc4a439d3832421c09ff21 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b12444adf9894901988a23249653e804 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ceaf06fe5574a438b7e8051b716fa95 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cceca86a9644ba0819cbcedf8758519 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9c734146474911b4baf19f960f436b : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/158b7a0e89bc47fb9058e446e6f745a1 : SUCCESS in 1h 02m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4df6c58931ab457bbac1d1b48d67dca7 : FAILURE in 1h 00m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aece8d595f04e86ad83f57bcae42130 : SUCCESS in 6m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/67a8277ad7f845a4962dc003eb8dbe1b : SUCCESS in 1h 04m 23s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fe4a5bd03dcb4a03b9a7593aea3cdcd6 : SUCCESS in 1h 11m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99adb90fee6b47aca1945215330bcd50 : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":4},{"id":"dbff5d0cc001461f121c6c68470bcf8cd20d34be","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-30 14:39:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"15c7c8625399ebf28e9d3ce91a98493283ee1eaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 15:03:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21277abecd2c4192b6161fcaba3188e7 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/faecefaa14e744ca8b94eabb65f66b56 : SUCCESS in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"88e4b3dcf7229bf24dabe3f7c63e859d33b34f55","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-30 16:23:16.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/779436/4/iscsi/ : SUCCESS in 1h 42m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/779436/4/smb/ : FAILURE in 1h 31m 41s","accounts_in_message":[],"_revision_number":4},{"id":"23491b7f3aa77597a765d9baecb46415878b1a3d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-30 16:36:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/36/779436/4/check/dsvm-tempest-zadara-driver/e54b825 : SUCCESS in 1h 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c966a40d45a467af870a7ba843fa5dff0f20042d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 17:04:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9439848a5f446709516f788e47333d5 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e5abaab5164947a89002fcf4f0b8d6 : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a77ede7f58a4ac9b3323b6c14a02395 : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e08f961bc1124581b4c2bf3cc5340607 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9529ab67e4c4912a3db1c1893b70eb8 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf8444764d6417599abd1a9e857cddf : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/81aa5485b3bb4ccd89ac87b901441737 : SUCCESS in 58m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50adbeede6fa474080b720fe8c927167 : SUCCESS in 1h 22m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/233cb53c0fe14e6494516f1d2ba22a94 : SUCCESS in 7m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f76f0d757c944ab2a9d1707ebe5dcd34 : SUCCESS in 12m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7776cb8e19324dc58146d725a93758ff : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/601ef9d3a02a431b8476357dcab564a9 : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7eddf64dcf24c4ea7e9ce414fa83cfa : SUCCESS in 8m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e14dd5bd0499405a8ef7f843a9bd4ff4 : SUCCESS in 37m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cc116756fa44d1796865fdbc63e45df : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ade7217a7b6a43b689d2733725fc551c : SUCCESS in 1h 09m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1d66b53bd15844668e6299e21df7a96f : FAILURE in 1h 27m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9806f458c50d42bd80ffe31ca5b59b18 : SUCCESS in 1h 39m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/37667a533b4a48709a80b5e1150904c6 : SUCCESS in 1h 13m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/906cd59cba584138869fbc61d07fd163 : SUCCESS in 1h 33m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d751e8266c2c4179aada6dae3a1bbdab : SUCCESS in 1h 24m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3901eaea8a2f4b5699f73dd512bb134c : SUCCESS in 2h 17m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df708a264ae346a7bd9077d78b787bdc : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":4},{"id":"82c2c8fa4e5304bc77ace04a92846617d4ec98c0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-30 17:05:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/36/779436/4/upstream-check/cinder-cDOT-iSCSI/05bb2cb/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/36/779436/4/upstream-check/cinder-cDOT-NFS/da58be1/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/36/779436/4/upstream-check/cinder-cDOT-FCP/ea522a3/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"09194fd8a264162dcd0136c2afd8c9ec1dd86d5b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-30 17:05:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/36/779436/4/upstream-solidfire-check/cinder-solidfire-iSCSI/21d4b78/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"74825b3a776255ba7e3b881204666f4ec76936a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-30 17:20:35.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a51fc96f75ac494f90c3f5ff3aadaffa : POST_FAILURE in 1h 34m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7efed7d96d974f3d98f95123bdd4605d : POST_FAILURE in 1h 40m 55s","accounts_in_message":[],"_revision_number":4},{"id":"502b8040dd6abf755fb51e478b2e3441fc7e6831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 17:32:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d41c31e5692bcb65e7701054b763efb0a938f874","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-30 17:37:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/779436/4/check/dsvm-tempest-cisco-zonemanager-job/8b4fe4c/ : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d11701ee9b6119343215063a093c2b7447cbd9d7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-30 17:54:17.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerStore_iSCSI/6f6356c/DellEMC_PowerStore_iSCSI/31 : SUCCESS in 48m 37s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerStore_FC/b8c0e26/DellEMC_PowerStore_FC/32 : SUCCESS in 47m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"62863ec9bd841534c315af5e745ef7948071f1b5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-30 18:31:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/779436/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3c1c3d5/ : SUCCESS in 2h 24m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/779436/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/151b604/ : SUCCESS in 1h 54m 04s","accounts_in_message":[],"_revision_number":4},{"id":"36a736f20df87420a2e2ec3b0ef17f0f60660ae2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-30 19:30:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/36/779436/4/check/cinder-storpool-tempest/29b0287/ : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":4},{"id":"1f708a58ea54d9c8bea2790e0d82830f6215c508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 20:19:28.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c3b1a82af64b6d5b8da286955071acc6657a3d13","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-30 23:07:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_SC_FC/18f9de9/DellEMC_SC_FC/54 : FAILURE in 1h 53m 17s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_SC_iSCSI/fbcb3ad/DellEMC_SC_iSCSI/51 : SUCCESS in 2h 07m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"b76251ec03a032149373de9ee167a8d8eb64455c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-01 00:16:59.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/779436-4/Seagate-iSCSI : SUCCESS in 46:14.71\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/779436-4/Seagate-FC : SUCCESS in 41:03.72","accounts_in_message":[],"_revision_number":4},{"id":"812a655d31b2ad12c666b69dba9d1a38b1d28ec3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-01 01:02:25.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/779436-4/Lenovo-iSCSI : SUCCESS in 49:48.66\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/779436-4/Lenovo-FC : SUCCESS in 47:03.25","accounts_in_message":[],"_revision_number":4},{"id":"a0ab4d8b32d4d73d2fcb0391b510b223d460689d","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-01 01:26:47.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/779436-4/HPMSA-iSCSI : FAILURE in 1:05:01\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/779436-4/HPMSA-FC : SUCCESS in 34:30.65","accounts_in_message":[],"_revision_number":4},{"id":"2c357b7bfae4b9596318484f844efdd712ff1a4d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-01 02:10:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_Unity_FC/eed9d8d/DellEMC_Unity_FC/90 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_Unity_iSCSI/0527d5f/DellEMC_Unity_iSCSI/84 : SUCCESS in 1h 27m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8cdd8bc945f1c778412840e0d2bfd850157dc7fd","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-05-01 02:59:02.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/779436-4/PVME-iSCSI : SUCCESS in 1:03:23\n* PVME-FC http://os-logs.tristero.net/dell-ci/779436-4/PVME-FC : SUCCESS in 56:17.94","accounts_in_message":[],"_revision_number":4},{"id":"3e4122ab19a5d67466402b66ced16e0504d881ad","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-01 03:17:56.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_VNX_iSCSI/676f59f/DellEMC_VNX_iSCSI/58 : SUCCESS in 3h 13m 57s\n- DellEMC_VNX_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_VNX_FC/41b52cd/DellEMC_VNX_FC/61 : SUCCESS in 1h 10m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"a4a817866887d5e588f099132eee50e8ddb30b67","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-01 06:31:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerFlex/9c4061e/DellEMC_PowerFlex/33 : ABORTED in 4h 00m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"cd9c462e784ff9b546f0a1cbc18d82e55e504496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 07:54:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7750ce023250461685b65416220e263d : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c89370247d4ccb98f668a57d83a65d : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24ab44090d1341549028c461d152e7c6 : SUCCESS in 11m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b7a65d04b674a7aa59e54ce2030ebfc : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd8bc4889f74980841593034a7f80c6 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/74b2320768d14ea884eb6fb3a7c63256 : SUCCESS in 54m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/43eae33f1642445ab5d4016578e1c10c : SUCCESS in 1h 07m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6027eccab2074b4e8f0a64c24c32d038 : SUCCESS in 7m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a521b3467064ad9ad25a3994ac92dc1 : SUCCESS in 1h 15m 03s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a473fb6558fa4ff9a53558454ddbfaf9 : SUCCESS in 58m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08b714cc54a2492dadb32a90a6f8bc42 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":4},{"id":"db28eb1ebc09ea29158af0db79845dd6206a0ed1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 07:54:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2b07346e19d6b9fe16bdbf253a1e0fc9b1c4c004","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 07:57:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57dd1a2e556e4c209543634c216df3e5 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74a3e54e888a4015a620ac99666f2446 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4},{"id":"9dfb6518cb97026db3acc58622666c5718a0baa5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-01 09:57:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_XtremIO_FC/762eea6/DellEMC_XtremIO_FC/103 : SUCCESS in 2h 08m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_XtremIO_iSCSI/8c882b7/DellEMC_XtremIO_iSCSI/103 : SUCCESS in 1h 53m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"c505325bc110bafc5eb36511478c21380561e8e9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-03 02:15:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerMAX_FC/25c3039/DellEMC_PowerMAX_FC/48 : SUCCESS in 4h 46m 31s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/36/779436/4/check/DellEMC_PowerMAX_iSCSI/705c2df/DellEMC_PowerMAX_iSCSI/42 : FAILURE in 9h 53m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b5b681b7371b261434ea66204b584f4c6edc03f0","revisions":{"9b7a3df7e476aa9f51eebca84cb671d00c24ee31":{"kind":"REWORK","_number":1,"created":"2021-03-09 08:07:06.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/36/779436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/779436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/779436/1"}}},"commit":{"parents":[{"commit":"a4e64d839fae2faedb54a47ccaa4e55175a39756","subject":"Merge \"Simplify composite check strings for project personas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4e64d839fae2faedb54a47ccaa4e55175a39756"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-03-09 08:05:35.000000000","tz":-480},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-03-09 08:07:03.000000000","tz":-480},"subject":"Bug fix for free space calculation","message":"Bug fix for free space calculation\n\nChange-Id: Iff02775822056904e5457ecddcefb2e580f47953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b7a3df7e476aa9f51eebca84cb671d00c24ee31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b7a3df7e476aa9f51eebca84cb671d00c24ee31"}]},"branch":"refs/heads/master"},"2676ca9a8db932e33d1aa97b527ff6758664e1f3":{"kind":"REWORK","_number":2,"created":"2021-03-25 05:53:39.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/36/779436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/779436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/779436/2"}}},"commit":{"parents":[{"commit":"47cdd944b3d134c62f4d8774a12d3e068031806b","subject":"Merge \"Support mTLS when calling the glance API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47cdd944b3d134c62f4d8774a12d3e068031806b"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-03-09 08:05:35.000000000","tz":-480},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-03-25 05:53:37.000000000","tz":-420},"subject":"Bug fix for free space calculation","message":"Bug fix for free space calculation\n\nFree space can be calculated from the storage array information\n\nCloses-bug: #1918229\nChange-Id: Iff02775822056904e5457ecddcefb2e580f47953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2676ca9a8db932e33d1aa97b527ff6758664e1f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2676ca9a8db932e33d1aa97b527ff6758664e1f3"}]},"branch":"refs/heads/master"},"399ccd80cb80000d70b6162243c6176858e142e2":{"kind":"REWORK","_number":3,"created":"2021-04-09 14:58:34.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/36/779436/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/779436/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/779436/3"}}},"commit":{"parents":[{"commit":"47cdd944b3d134c62f4d8774a12d3e068031806b","subject":"Merge \"Support mTLS when calling the glance API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47cdd944b3d134c62f4d8774a12d3e068031806b"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-03-09 08:05:35.000000000","tz":-480},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-09 14:56:27.000000000","tz":-420},"subject":"Bug fix for free space calculation","message":"Bug fix for free space calculation\n\nInstead of calculating the free space ourselves, simply report the amount of\nfree space that the storage array tells us it has.\n\nCloses-bug: #1918229\nChange-Id: Iff02775822056904e5457ecddcefb2e580f47953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/399ccd80cb80000d70b6162243c6176858e142e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/399ccd80cb80000d70b6162243c6176858e142e2"}]},"branch":"refs/heads/master"},"b5b681b7371b261434ea66204b584f4c6edc03f0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-04-09 23:15:30.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/36/779436/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/779436/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/779436/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/779436/4"}}},"commit":{"parents":[{"commit":"47cdd944b3d134c62f4d8774a12d3e068031806b","subject":"Merge \"Support mTLS when calling the glance API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47cdd944b3d134c62f4d8774a12d3e068031806b"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-03-09 08:05:35.000000000","tz":-480},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-09 23:13:08.000000000","tz":-420},"subject":"Bug fix for free space calculation","message":"Bug fix for free space calculation\n\nInstead of calculating the free space ourselves, simply report the\namount of free space that the storage array tells us it has.\n\nCloses-bug: #1918229\nChange-Id: Iff02775822056904e5457ecddcefb2e580f47953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b681b7371b261434ea66204b584f4c6edc03f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b681b7371b261434ea66204b584f4c6edc03f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
