)]}'
{"id":"openstack%2Fcinder~780018","triplet_id":"openstack%2Fcinder~master~I743e9999a42c6a7a978c88b16f02b7c5e6e826c7","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"I743e9999a42c6a7a978c88b16f02b7c5e6e826c7","subject":"Fix invalid yaml in reno template","status":"MERGED","created":"2021-03-11 15:41:49.000000000","updated":"2021-04-09 17:02:22.000000000","submitted":"2021-04-09 17:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780018","meta_rev_id":"0dd6d5cc4eb71b52d0779661878ae6f69e444258","_number":780018,"virtual_id_number":780018,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 16:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-04-09 14:04:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-15 07:09:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-11 15:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-04-09 14:04:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 15:43:17.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-03-11 16:58:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 18:48:49.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-11 20:19:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 23:56:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-12 00:03:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 00:24:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-12 01:21:57.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-12 10:36:22.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-12 12:18:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 13:50:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 16:37:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 19:39:01.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 20:40:48.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 21:44:56.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-13 05:52:20.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-13 12:02:55.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-14 02:53:01.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-15 07:09:13.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-03-15 15:52:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 14:04:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"611b158c5338bbfa34b00fa714b62687145d3ae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-11 15:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd16875f05bb0388310086d70a37e5b8b1d8bc91","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-11 15:43:17.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"a4f9a1f871d65fc1b98a1eb8660d899762d7e1e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-11 16:58:06.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/780018/780018/1/check-not-vote/ext-cinder-zuul/cf10d3b : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8b0ea83b4ec5e2bbce456ee9ed6d213b0564b81f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-11 18:48:49.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/18/780018/1/check/dsvm-tempest-TOYOU-ACS5000-driver/6518d2a : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4b40d7df7a25757b1c4aae7710612a81a9e41971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 20:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8e342445b94f348d159ae4a225f5c2 : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372cfc97f8a64c71a1dcdfd9e59bb5f8 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97d382ded90a450abfc686d0bd5f0d47 : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15683036e97a48da934f6533f06e9df3 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb128f3e970416c948efbd7c39a57ff : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b65dcd232c4a71a01d647ef4855f40 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b0c2b244fee640b69c1971a873e676c9 : SUCCESS in 1h 15m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/307de90de4e84e858240f13f83650f32 : SUCCESS in 1h 27m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9abc7e4035714f868eeb4fc7fc593be1 : SUCCESS in 11m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2ab3e6cb8f204805be97c7dfa18c486f : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6e14946f9ac470886953a0d4a5aaa65 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb80f773da6b480b86bcea9472cb3dac : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d5058d9bdd024b1c82f18c19ffbb57f2 : SUCCESS in 42m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42a2bd554dda4125a63a36acd4547081 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e207925f7cea41f4a33aac8c50c68b33 : SUCCESS in 1h 33m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40a77de0e5c44aae91fcb407a3035217 : FAILURE in 1h 23m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/028bb9b925814c7cb0c0eace70f2c428 : SUCCESS in 1h 23m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c72df6b58179495ba3481a31b5071e72 : SUCCESS in 1h 10m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d65637e3dfd44cf997abcdb781400018 : SUCCESS in 1h 23m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c99190466e0f40f5b6879c02f249f3cf : SUCCESS in 1h 29m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89ff0387da7b4e7e922a152b550306fc : SUCCESS in 2h 02m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/592e3c55f8f6486f9d0b12a764a81f3f : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7c12e43042d8f27222e5362cdbeee8e6b7d0d21d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-11 23:56:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/18/780018/1/check/DellEMC_PowerStore_iSCSI/f74104d/DellEMC_PowerStore_iSCSI/580 : SUCCESS in 56m 07s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/18/780018/1/check/DellEMC_PowerStore_FC/b6718e4/DellEMC_PowerStore_FC/574 : SUCCESS in 53m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"7fd168b890c1de20665c4efac71985ad185f2451","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-12 00:03:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/18/780018/1/check/DellEMC_PowerFlex/3c94c03/DellEMC_PowerFlex/548 : FAILURE in 57m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"0865360a11384468700b54da99017135f96726dd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-12 00:24:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/18/780018/1/check/dsvm-tempest-cisco-zonemanager-job/5dae223/ : FAILURE in 1h 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9d5f6518742dafe393cbe7e8fdf9d3328be4de7d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-12 01:21:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/780018/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/23c69a2/ : SUCCESS in 2h 21m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/780018/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/56dcd93/ : SUCCESS in 2h 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"beda12c91a348f48b66346e012d456042094682f","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-12 10:36:22.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/780018-1/Seagate-iSCSI : SUCCESS in 44:32.03\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/780018-1/Seagate-FC : SUCCESS in 44:01.50","accounts_in_message":[],"_revision_number":1},{"id":"4c584962f9a0a1a1668f8ea72cab72a0a582938d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-12 12:18:24.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/780018/1/check/tempest-dsvm-full-focal-py3/ac78526/ : SUCCESS in 1h 59m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/780018/1/check/devstack-plugin-ceph-dsvm-focal-py3/e75790e/ : TIMED_OUT in 2h 38m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b57369bf461ac1135400f20f29aeaa4d23660fff","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-12 13:50:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/780018/1/ : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ca925c5c63618452f7993b1c0566d3a7f80a6a68","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-12 16:37:36.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-18-780018-1/ns5_nfs : FAILURE in 4h 14m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-18-780018-1/ns5_iscsi : FAILURE in 4h 17m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"22bc649d35db0c226357ad9e7d0005189e32b9ce","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-12 19:39:01.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/18/780018/1/check/DellEMC_VNX_iSCSI/13d1bce/DellEMC_VNX_iSCSI/542 : SUCCESS in 3h 19m 56s\n- DellEMC_VNX_FC http://publiclogs.emc.com/18/780018/1/check/DellEMC_VNX_FC/14d6847/DellEMC_VNX_FC/569 : SUCCESS in 1h 58m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3dc26eea2966321cc5da013149fca2a4aca9470b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-12 20:40:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/18/780018/1/check/dsvm-tempest-zadara-driver/ae03c83 : SUCCESS in 1h 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62a1375d646903f9f00fbdbce6e2f19d64151267","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-12 21:44:56.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/780018-1/HPMSA-iSCSI : FAILURE in 1:03:38\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/780018-1/HPMSA-FC : SUCCESS in 36:35.74","accounts_in_message":[],"_revision_number":1},{"id":"31ffc291a34a175ab0dc4a71945429250332d4f1","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-13 05:52:20.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/780018-1/Lenovo-iSCSI : SUCCESS in 49:11.39\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/780018-1/Lenovo-FC : SUCCESS in 46:35.70","accounts_in_message":[],"_revision_number":1},{"id":"5c9cea3ca3c1cec0d0b1f0cc1bad0d8fbfb88ce3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-13 12:02:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/18/780018/1/check/DellEMC_XtremIO_FC/cfbe2b5/DellEMC_XtremIO_FC/981 : FAILURE in 1m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/18/780018/1/check/DellEMC_XtremIO_iSCSI/45ef48d/DellEMC_XtremIO_iSCSI/986 : SUCCESS in 1h 44m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"18dfb78a507b843daf005ed2185c31f0462573ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-13 19:20:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-18-780018-1/ns5_nfs : SUCCESS in 6h 47m 10s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-18-780018-1/ns5_iscsi : SUCCESS in 6h 40m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ebda7321db6dbf934acb6ebb79fbe399e677f9c4","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-14 02:53:01.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/780018-1/PVME-iSCSI : SUCCESS in 57:38.40\n* PVME-FC http://os-logs.tristero.net/dell-ci/780018-1/PVME-FC : SUCCESS in 56:15.10","accounts_in_message":[],"_revision_number":1},{"id":"3e510a088a00f2ac0445ca92b5eab61546a68115","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-15 07:09:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0889efaac7c8b94e15a018eeaaf63d1a1bbf7804","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-15 15:52:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/780018/1/20210315221621/NEC-ISCSI : SUCCESS in 2h 24m 46s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/780018/1/20210315221621/NEC-FC : SUCCESS in 2h 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b19f3c06e5a19814d97b512bc79ac2f442e2d38b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-09 14:04:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, Eric.","accounts_in_message":[],"_revision_number":1},{"id":"677fbf23210e0a109e68e25bfad2f69b76650bd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 14:05:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0543ce0f978f0f59feafa280b1a0a4798f788f00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 15:26:15.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e15226a2733a877c4a95bbe34f1e1485c09a5772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 16:59:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5439018f771343e4a8f1ade8fe1287a2 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19218792b23f4fa588b1c1c7b2a9e69e : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd7c62c98946401cb07e1e852717e85c : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/401cc837030041009706eec5d88a1259 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f256d6f0c314c528584f58f6f9a5f06 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/8a73d13711b74fa384afe5fcb2f08eba : SUCCESS in 49m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e8e465ed70f42e697db38e39722dbef : SUCCESS in 1h 06m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bcd68de8ed3045e0bfaa368612ade2d1 : SUCCESS in 1h 07m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13e13cb924654740a10f07cb9c3cb637 : SUCCESS in 1h 32m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51222bf8cff64aeab3173f779af531c9 : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":1},{"id":"555d4462e10859621d3da6d20ca7c4754b48f0b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:00:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0dd6d5cc4eb71b52d0779661878ae6f69e444258","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:02:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b6f73a10b647a9acd00668f81c94c5 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f84fff321a0c9d9f3c69ae8110c2e0413fbc870","revisions":{"7f84fff321a0c9d9f3c69ae8110c2e0413fbc870":{"kind":"REWORK","_number":1,"created":"2021-03-11 15:41:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/18/780018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/780018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/780018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/780018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/780018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/780018/1"}}},"commit":{"parents":[{"commit":"a4e64d839fae2faedb54a47ccaa4e55175a39756","subject":"Merge \"Simplify composite check strings for project personas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4e64d839fae2faedb54a47ccaa4e55175a39756"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-11 15:35:46.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-11 15:35:46.000000000","tz":-300},"subject":"Fix invalid yaml in reno template","message":"Fix invalid yaml in reno template\n\nThis fixes two issues with the reno yaml template.\n\nThe prelude paragraph has a line longer than 80 chars.\n\nAdd a newline for the bug # entry.  This prevents an error\nwhere yamllint complains that a token cannot start with\na backtick, if the bug # entry is used as the first item\nin the \"fixes\" section.\n\nChange-Id: I743e9999a42c6a7a978c88b16f02b7c5e6e826c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f84fff321a0c9d9f3c69ae8110c2e0413fbc870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f84fff321a0c9d9f3c69ae8110c2e0413fbc870"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
