)]}'
{"id":"openstack%2Fcinder~780624","triplet_id":"openstack%2Fcinder~master~I01bdb2d2bd74a8b236b7e71471759e1a8fd59511","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"I01bdb2d2bd74a8b236b7e71471759e1a8fd59511","subject":"Fix tox -e pep8","status":"ABANDONED","created":"2021-03-15 17:06:42.000000000","updated":"2021-03-17 15:12:44.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"fef2685d5ee132997e44ff09f25083a852c245a2","_number":780624,"virtual_id_number":780624,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-15 18:41:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2021-03-16 03:38:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2021-03-15 17:06:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 17:12:12.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-15 17:51:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 18:35:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-15 18:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 19:56:37.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 23:33:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 23:40:41.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 01:00:26.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-16 01:53:48.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b025ab39afbe4e7bc16b232e355555ea28b98360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-15 17:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bd979c47ae9d31ed3afeb876b7ce244fbe616d6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-15 17:06:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"015de14f02046d87560f54721f4141054d3f108c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-15 17:12:12.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/24/780624/1/check/dsvm-tempest-TOYOU-ACS5000-driver/885a299 : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ded61750e404ae4e6071b8c47c999f0f154b1bc7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-15 17:51:38.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/780624/780624/1/check-not-vote/ext-cinder-zuul/73ae60c : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"85f94935696f4d4261e93e828432c8c050989826","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-15 18:35:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/780624/1/check/dsvm-tempest-cisco-zonemanager-job/a357ed2/ : FAILURE in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"034b9607d1779c1bdd15c52f0c01151db555f38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:41:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6f77e79b04d64371ac3a80e99966b703 : SUCCESS in 11m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1bf8bacec34bd98d94ec21b67deed6 : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163e48f925e0492492de6ba4d73e3635 : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e311e13d05f4ef497b0e28efdcc3c7d : FAILURE in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dfa7fa6795944a6b6bef8db55678d07 : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/491a35516e514f9ba4a1f57b380d736d : FAILURE in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8c5b27eac54a30b0df371017bc1044 : FAILURE in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a73c742755944c4e9e30db399a949be2 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c80738f5346457f9421c7b42d01681c : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acbc374652eb43ceb3d96b3133a09bfa : FAILURE in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f114b76d30ba48108b0d6ade7d3ded00 : FAILURE in 4m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7d931912dfcb44d489edd2142bd182c2 : SUCCESS in 48m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a210647cb5c9406fbf5929e34b7bd4e4 : TIMED_OUT in 1h 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2affa09a8b5f471033440c8873e79ea6e0e7f7a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-15 19:56:37.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/3df75ddaa97b4ae0a59ab7d6c2fbd36a : RETRY_LIMIT in 30s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/24/780624/1/check/devstack-plugin-ceph-dsvm-focal-py3/f1090ab/ : FAILURE in 2h 05m 32s","accounts_in_message":[],"_revision_number":1},{"id":"84637cc06ef0774b643ba88bb6baf19dabbc4b6b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-15 23:33:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-780624-1/ns5_nfs : SUCCESS in 5h 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-780624-1/ns5_iscsi : SUCCESS in 4h 54m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c873b61372b1b34080850ccdb75140c726f18b3e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-15 23:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/24/780624/1/check/dsvm-tempest-zadara-driver/e0ad207 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e448cec885e1411b76e6d2049cbe19920ed0a501","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-16 01:00:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nSee alternative suggestion inline.","accounts_in_message":[],"_revision_number":1},{"id":"b3cca83ed50f553fc07e914664ab8a6b0bcaa291","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-16 01:53:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/24/780624/1/check/DellEMC_XtremIO_FC/0379173/DellEMC_XtremIO_FC/1017 : SUCCESS in 2h 02m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/24/780624/1/check/DellEMC_XtremIO_iSCSI/8530423/DellEMC_XtremIO_iSCSI/1019 : SUCCESS in 1h 43m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"a43cdd3aa3c07c74f36d54af0e9df426bbb156f9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-16 03:38:53.000000000","message":"Patch Set 1: -Code-Review\n\nI went ahead and put up a patch for my alternative suggestion: https://review.opendev.org/c/openstack/cinder/+/780692","accounts_in_message":[],"_revision_number":1},{"id":"fef2685d5ee132997e44ff09f25083a852c245a2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-17 15:12:44.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/cinder/+/780692","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aac60611416b180291f2467dbd4ce8d6a9b04ff9","revisions":{"aac60611416b180291f2467dbd4ce8d6a9b04ff9":{"kind":"REWORK","_number":1,"created":"2021-03-15 17:06:42.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/24/780624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/780624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/780624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/780624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/780624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/780624/1"}}},"commit":{"parents":[{"commit":"a66f688992dae0a7bd26cdfc866af539be15ebca","subject":"Merge \"Support format info in fs type drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a66f688992dae0a7bd26cdfc866af539be15ebca"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-15 17:04:55.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-15 17:06:32.000000000","tz":-240},"subject":"Fix tox -e pep8","message":"Fix tox -e pep8\n\npycodestyle 2.7.0 causes flake8 to break unless\nusing a sufficiently new version of flake8.\n\nChange-Id: I01bdb2d2bd74a8b236b7e71471759e1a8fd59511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aac60611416b180291f2467dbd4ce8d6a9b04ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aac60611416b180291f2467dbd4ce8d6a9b04ff9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
