)]}'
{"id":"openstack%2Fcinder~780829","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~I29bd3b74e8864f7ae02904a5402db2afe2dbd94b","project":"openstack/cinder","branch":"stable/victoria","hashtags":[],"change_id":"I29bd3b74e8864f7ae02904a5402db2afe2dbd94b","subject":"NetApp SolidFire: Refactor DuplicateSfVolumeNames exception","status":"MERGED","created":"2021-03-16 14:41:14.000000000","updated":"2021-04-09 17:05:56.000000000","submitted":"2021-04-09 17:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780829","cherry_pick_of_change":770051,"cherry_pick_of_patch_set":1,"meta_rev_id":"92800b36a5f660f156a2094ede0ed0f0d0a033d8","_number":780829,"virtual_id_number":780829,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 17:03:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2021-04-05 03:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":2,"date":"2021-04-09 15:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-19 15:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2021-04-09 15:40:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 15:05:50.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-16 15:22:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 17:08:59.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-16 17:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 18:15:11.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 19:06:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 19:21:22.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-16 19:28:33.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-17 05:11:38.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 14:19:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 17:40:41.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 15:39:17.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-04-05 03:12:36.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-04-09 15:40:18.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"702af2a3c77f877070823675d561ad439ce1ca2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-16 14:41:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"447e21e083c9e96295aa75a25b3b6ac2de2216e1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-16 15:05:50.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/29/780829/1/check/dsvm-tempest-TOYOU-ACS5000-driver/73edc3d : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"97077b3bfdbe7ac836c94a0a5ae3382b4833449b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-16 15:22:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/29/780829/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/19d028b/19d028b.html : FAILURE in 39m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a1742765d0d2f4ee6c2b35296066879674b5b7a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-16 16:44:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/29/780829/1/check/iscsi-driver-nimble/c311ae5/c311ae5.html : SUCCESS in 2h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"564b861a306b5133c0c9a23d9b74c536df733539","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-16 17:08:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/29/780829/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/88eb194/ : FAILURE in 29m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/29/780829/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7890494/ : FAILURE in 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"96dd67bc917aacb7c180c484980992af88c37c68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 17:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77334cb3e49c49ae9360528e4c7f11e8 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304782b765154051b2da3415192831b5 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78da2cfae44e4e6c85d58f54b481d322 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64cfe28f178d489fbec7accee858b865 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834c258642e045dc9f558b5c763908d0 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/c7e1b9863ee34170b037315e4db507ba : SUCCESS in 1h 01m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ed95fe4fab9496e923cc58cd4f88609 : SUCCESS in 1h 08m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d66b3e2b7f96433e994cabc5855769c1 : SUCCESS in 14m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/990821896459429cb17d287ae025820e : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35c4dff99a304ce48497d4846b1b20dd : SUCCESS in 8m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e23c206708a44ce5bb731706341eed1a : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6c3ea782f723436baea02505e90f5974 : SUCCESS in 43m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b9eb412239447ffaab4e566372c17ac : FAILURE in 7m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d541051e28304a05bd2c90ce8ba4dcdf : SUCCESS in 1h 29m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1220c4ef65ea4743a18db04e27fd1e85 : FAILURE in 1h 43m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9c6052e62f6149889ecf45106fac31ef : SUCCESS in 1h 57m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c4a92744a83483db98434478379a083 : SUCCESS in 1h 15m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/34a3f379006e4c9d8271aa1d310bca2a : SUCCESS in 1h 19m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1fa8e65aab3e4fcfb28b09ddb74c76e3 : SUCCESS in 1h 24m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82781569ce434c09a5ef0b2b27d8b449 : SUCCESS in 2h 11m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8a27a4750f748cc9ae211050e40eca6 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1333aa6a06602d7c8235446e40ee53db9bffe9c3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-16 18:15:11.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/29/780829/1/check/tempest-dsvm-full-focal-py3/ab8e8d0/ : SUCCESS in 1h 57m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/29/780829/1/check/devstack-plugin-ceph-dsvm-focal-py3/07588cc/ : FAILURE in 1h 09m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f1cc06cc24acdaada2993f8d32691c35245b4d4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-16 19:06:26.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/29/780829/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fc9df61696b0b2aaddb5220b72d3ec6abdc49068","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-16 19:21:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/29/780829/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f40e7b4/ : FAILURE in 1h 28m 44s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"fb5c8061c156af21b480f3247e6c5e8f77927998","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-16 19:28:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/29/780829/1/check/dsvm-tempest-cisco-zonemanager-job/b8b9ae3/ : FAILURE in 1h 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7cdc855594b3839b8d3a22bf84abab37ced852d9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-17 05:11:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/29/780829/1/check/dsvm-tempest-zadara-driver/1343a7c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b097c8bdf9e424373e7348a9b2d4d8ee79d4637","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-17 10:48:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/29/780829/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c62bbc095d24726bb45b48d3bf82b46f8ee161ef","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-17 14:19:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-29-780829-1/ns5_nfs : SUCCESS in 4h 42m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-29-780829-1/ns5_iscsi : SUCCESS in 4h 24m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f7d6075c71f6ec16c2fe9738389a2860ca1d024a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-17 21:25:46.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/29/780829/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"212cbceecbe1cd4bbd53261de66a3fde8af2a0e3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-18 17:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/29/780829/1/20210319001800/NEC-FC : SUCCESS in 2h 04m 55s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/29/780829/1/20210319001800/NEC-ISCSI : SUCCESS in 2h 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"685bbfda791677ab67fad8cc7d18870dd02d5aa1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-19 15:39:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLimited to the driver and it looks like an appropriate backport.","accounts_in_message":[],"_revision_number":1},{"id":"3e775014dd08bca2aa71f3bdbb84de8cbee32fcb","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-04-05 03:12:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4e720e2a48d075949c76e810cfad204991ae3a1e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-09 15:40:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e96e8c4feeab4888678266460086a906ab93817e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 15:40:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1e5a0f39f165545de39b11461d5e871b939b608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:03:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4ddf98a85924263a2e575f4bff1f474 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd7f028532442df93bb3f19787ea4b0 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b50a327451e54af28fb7a76e2802029a : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e06a0320ab60490fab21fc8793e0ef13 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b6cf12800f74ad1adc3bce9b231cf15 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/9c5bdbc13a57497f9b4e18a17b7a3976 : SUCCESS in 52m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f195e171516477d97c9656381fac5a9 : SUCCESS in 1h 17m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a0d3107b52394628b6f26cbf5d8effa9 : SUCCESS in 1h 05m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83b303da336341daac05cd27545d0cd3 : SUCCESS in 1h 18m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf6221b466374b52a3b247f7d6e071fc : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d2c8e98f86e468a1f331f7995bfe176a12e2705c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:03:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92800b36a5f660f156a2094ede0ed0f0d0a033d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:05:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33234553f7448d3bd294ba5ea133801 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cd1614f2ab641c526225ced5a9029250eb5050c","revisions":{"9cd1614f2ab641c526225ced5a9029250eb5050c":{"kind":"REWORK","_number":1,"created":"2021-03-16 14:41:14.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/29/780829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/780829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/780829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/780829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/780829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/780829/1"}}},"commit":{"parents":[{"commit":"7d18029d52398080525f67383bbdadfa0786c788","subject":"Pure Storage: check volumename length does not exceed maximum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d18029d52398080525f67383bbdadfa0786c788"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-12-30 01:41:40.000000000","tz":0},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-16 14:41:14.000000000","tz":0},"subject":"NetApp SolidFire: Refactor DuplicateSfVolumeNames exception","message":"NetApp SolidFire: Refactor DuplicateSfVolumeNames exception\n\nThis patch refactors the DuplicateSfvolumeNames exception to\naddress the following issues:\n\n- Inherits from exception.Duplicate which inherits from\nCinderException directly rather than from VolumeDriverException\n- Naming pattern differs from the other SolidFire exceptions\n- Exception message differs in parts of the code.\n\nChange-Id: I29bd3b74e8864f7ae02904a5402db2afe2dbd94b\n(cherry picked from commit 5e2d30bd431d111b9e69fb93ca1360efe5e078b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cd1614f2ab641c526225ced5a9029250eb5050c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cd1614f2ab641c526225ced5a9029250eb5050c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
