)]}'
{"id":"openstack%2Fcinder~781023","triplet_id":"openstack%2Fcinder~master~I1b8e4a6be38c6ffd0e15b05353ee4474de9f0212","project":"openstack/cinder","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"I1b8e4a6be38c6ffd0e15b05353ee4474de9f0212","subject":"remove six of cinder.test.unit.volume.drivers.vmware","status":"ABANDONED","created":"2021-03-17 06:28:11.000000000","updated":"2021-03-19 03:54:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c441ff54ce9bc5d5c3f2d5977e62186b9bed72e","_number":781023,"virtual_id_number":781023,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-17 07:48:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-2,"date":"2021-03-19 03:00:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 06:35:03.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-17 07:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 09:40:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 11:39:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 15:53:37.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-18 22:37:15.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-19 03:00:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-19 03:00:21.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 03:54:59.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"094a3ef59e840b8d5a13e67932dd3cacf9170970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-17 06:28:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ab66f75d770bf27da69c1cf181bec3aad02e66a","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-17 06:35:03.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/23/781023/1/check/dsvm-tempest-TOYOU-ACS5000-driver/602f7a3 : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"be32967f855356136304626f378cebfd605730d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-17 07:11:46.000000000","message":"Topic set to bp/six-removal","accounts_in_message":[],"_revision_number":1},{"id":"6a6837cad2d53362cfbca7135167a76cdbf31d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 07:48:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8636b19acb1b4e8baa33c27aa5586d66 : FAILURE in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eaa7f8e83284de7bc6e591bc3c1bf94 : FAILURE in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e94b7edfe624d4fa1ee04bdce9f2f45 : FAILURE in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eb0dd4cedd040a6ab1a84f11fe1a5f0 : FAILURE in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2334cf44eb04ea9b1330a13f773d6e7 : FAILURE in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d49bb081ddb41bb9fc37c89d7a7fdfb : SUCCESS in 11m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7a328c63e19499a9d9bcbbb31782c94 : FAILURE in 11m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37bdbc6a8d344073902d53104f149cbf : SUCCESS in 51m 26s (non-voting)\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_zadara.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_storpool.py","accounts_in_message":[],"_revision_number":1},{"id":"265d3a32c02f5d5757be6100e25c521ecdb01933","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-17 09:40:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/781023/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/77a398a/77a398a.html : FAILURE in 1h 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b1d677a76cac0aeda8bcff0a1bb6f5d5ffa3f1fa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-17 11:39:14.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/781023/781023/1/check-not-vote/ext-cinder-zuul/116261f : FAILURE in 4h 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"00d930b1567ba5f7c620fbf7635d1fdb65e67347","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-17 15:53:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/781023/1/check/dsvm-tempest-cisco-zonemanager-job/3eeabec/ : FAILURE in 1h 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ad8307d2577fe7000aa799123a03e56ce1752714","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-18 03:02:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/781023/1/check/iscsi-driver-nimble/497ef8c/497ef8c.html : FAILURE in 2h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a175fcc1cc0b9996997f52253e8896a2e7a6fda2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-18 22:37:15.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/781023/1/check/DellEMC_Unity_FC/eecdd12/DellEMC_Unity_FC/887 : SUCCESS in 1h 23m 26s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/781023/1/check/DellEMC_Unity_iSCSI/b58009c/DellEMC_Unity_iSCSI/839 : SUCCESS in 1h 56m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6bcc6a4f6134518a0d53eae223a1c00090aceb6d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-19 03:00:03.000000000","message":"Patch Set 1: Code-Review-2\n\nCinder project policy is that drivers are allowed to use six while stable branches that support py2 are open for backports.","accounts_in_message":[],"_revision_number":1},{"id":"fa9f00bb4b5720175b14a9f92bb8fc68d743470a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-19 03:00:21.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/781023/1/check/DellEMC_XtremIO_FC/7c586f9/DellEMC_XtremIO_FC/1052 : SUCCESS in 2h 04m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/781023/1/check/DellEMC_XtremIO_iSCSI/e445ec1/DellEMC_XtremIO_iSCSI/1056 : SUCCESS in 1h 45m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"44b5246ac47e3026449e526230e74f62c2e9a06f","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-19 03:04:26.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-2\n\u003e \n\u003e Cinder project policy is that drivers are allowed to use six while stable branches that support py2 are open for backports.\n\nok, got it, thanks very much","accounts_in_message":[],"_revision_number":1},{"id":"9e34ee02c8e4495aa85de7b74cda541b3a7aa4ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-19 03:04:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"5c441ff54ce9bc5d5c3f2d5977e62186b9bed72e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-19 03:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-781023-1/ns5_nfs : SUCCESS in 5h 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-781023-1/ns5_iscsi : SUCCESS in 4h 50m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54f34b8b4563ce5a774b9ec691ad0f110062f2ce","revisions":{"54f34b8b4563ce5a774b9ec691ad0f110062f2ce":{"kind":"REWORK","_number":1,"created":"2021-03-17 06:28:11.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/23/781023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/781023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/781023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/781023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/781023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/781023/1"}}},"commit":{"parents":[{"commit":"3dd437319df0797b44c81c8ccc589f437b9200ef","subject":"Remove six from cinder.tests.unit.volume.drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3dd437319df0797b44c81c8ccc589f437b9200ef"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-17 06:27:48.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-03-17 06:27:48.000000000","tz":480},"subject":"remove six of cinder.test.unit.volume.drivers.vmware","message":"remove six of cinder.test.unit.volume.drivers.vmware\n\n- six.text_type\n\nImplements: blueprint six-removal\nChange-Id: I1b8e4a6be38c6ffd0e15b05353ee4474de9f0212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54f34b8b4563ce5a774b9ec691ad0f110062f2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54f34b8b4563ce5a774b9ec691ad0f110062f2ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
