)]}'
{"id":"openstack%2Fcinder~781105","triplet_id":"openstack%2Fcinder~master~I099d101a3f2bcfbd2026306081a9858c610ab1ff","project":"openstack/cinder","branch":"master","topic":"unused-quota-classes","hashtags":[],"change_id":"I099d101a3f2bcfbd2026306081a9858c610ab1ff","subject":"Update cinder manage quota commands","status":"MERGED","created":"2021-03-17 14:00:25.000000000","updated":"2021-04-06 17:51:42.000000000","submitted":"2021-04-02 08:08:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781105-unused-quota-classes","meta_rev_id":"20d68b8a3266982167aae2ed5879472821416b83","_number":781105,"virtual_id_number":781105,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-02 08:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-03-31 16:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-31 14:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-31 16:58:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31305,"name":"zhubingyang","email":"yang77498441@126.com"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 16:41:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 17:25:18.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-17 20:34:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-17 20:38:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 21:41:02.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 23:33:38.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-18 04:07:31.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 05:14:10.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-18 07:05:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-18 14:01:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-19 01:13:36.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-20 01:02:32.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-20 04:20:41.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 08:35:51.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 03:00:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-21 11:03:24.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 01:00:59.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 01:25:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-24 17:48:32.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-26 19:33:06.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-27 13:18:38.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-29 07:00:52.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-29 22:08:55.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-31 16:58:32.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-03-31 20:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-04-04 14:37:52.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-04 15:23:48.000000000","updated_by":{"_account_id":31305,"name":"zhubingyang","email":"yang77498441@126.com"},"reviewer":{"_account_id":31305,"name":"zhubingyang","email":"yang77498441@126.com"},"state":"CC"}],"messages":[{"id":"19c7bdb25cbbd3dc814522045d47be2d43576383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-17 14:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdb4be89659c569e978c7a32ed97584ed62cb69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:41:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c437cf6bb8614722bdb4117b37d99f83 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f460e2441246f58ae6ae4832e659e9 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c3b9e2ba4941daa27c61c31a073565 : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ed1cd5eba8d486a8e314f3ce0210340 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27766e55a11b4f8f8628c5e13be759a4 : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef93cabc6a544e5f80e9cecce2018feb : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/045fcc154bc043fc80de4d3365a18e6e : SUCCESS in 49m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e297b802d5c4d8e94de86b564f01efd : SUCCESS in 1h 14m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/36b7684cf20a42fabb5f5c6bdefca365 : SUCCESS in 16m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0420878bafeb486ba5020710e9f68c31 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e147e90d56c4cd5a259a662c4f011c4 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/734e5383d35b428793e36c7516a29c63 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae60cfe02cdb4a1b879300d2c8cbdd8f : SUCCESS in 56m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d2121c9d9794457afdd48a4dbfbd1dd : FAILURE in 7m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5826540319654b0991e752fbeb87f560 : SUCCESS in 1h 20m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0c51d96c9dd048c6a45ace6aed59e635 : FAILURE in 1h 41m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/32d616f67e5f42e3b86fbf9e588be514 : SUCCESS in 1h 46m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/763937a6bcd547d9aeddc835ff7202ac : SUCCESS in 1h 11m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cace765538624cd994e803f2f0966bc4 : SUCCESS in 1h 37m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7df0cf682cbb427cb027d3d4359de08b : FAILURE in 1h 22m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3879dc80c5e84eb0926011566aceb4c7 : SUCCESS in 2h 23m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b726ca90bdf74514bdd60f1336dddb7f : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":1},{"id":"82e87d56c06b5b2aba2b2be5a11141a31dd169dc","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-17 17:25:18.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/05/781105/1/check/dsvm-tempest-TOYOU-ACS5000-driver/4bf7295 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b6d27edca012ae22e2e87655afa7e21f3fb32d66","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-17 20:34:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/781105/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f335ee6/ : FAILURE in 3h 49m 04s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"f2b8b0ca07374e45bc9200bb6b72adde9478ff4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-17 20:38:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c8d16bf3a907d4cfcc9388e242a382bd88d3cdb4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-17 21:41:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/781105/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6c195f7/6c195f7.html : FAILURE in 38m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6a006fe97530802aec72595ff96cebd023437953","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-17 23:33:38.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781105-1/HPMSA-iSCSI : FAILURE in 59:03.04\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781105-1/HPMSA-FC : SUCCESS in 34:24.93","accounts_in_message":[],"_revision_number":1},{"id":"62c63ad5ecb4fdbe475de16225b2f1028b5c3c53","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-18 04:07:31.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/eabbfc9596864d008857472d03c9a4e2 : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/781105/1/check/devstack-plugin-ceph-dsvm-focal-py3/00fc50c/ : SUCCESS in 2h 17m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b24ab9ec4d71e1e050a95160b29230789c6d629b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-18 05:14:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/05/781105/1/check/DellEMC_PowerStore_iSCSI/d3dcc7c/DellEMC_PowerStore_iSCSI/640 : SUCCESS in 52m 52s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/05/781105/1/check/DellEMC_PowerStore_FC/35a529b/DellEMC_PowerStore_FC/634 : SUCCESS in 51m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"0b3ad514daa020584a8c7f2e17050c898d63a89b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-18 07:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/781105/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a4e4d7f/ : SUCCESS in 2h 26m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/781105/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6356bbf/ : SUCCESS in 2h 22m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f2acf39a802e6bd7c9f3c4dd0709a45f196ffc5c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-18 09:35:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d3f11262af6014a175a5cb4579e362ff777f42bd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-18 14:01:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/781105/1/check/dsvm-tempest-cisco-zonemanager-job/b58a1ad/ : FAILURE in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fbf19d1aae97d3c7e83df2004b2e7a70680311dd","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-19 01:13:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/781105/1/check/DellEMC_SC_FC/92059d1/DellEMC_SC_FC/455 : SUCCESS in 1h 54m 41s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/781105/1/check/DellEMC_SC_iSCSI/aaafd0a/DellEMC_SC_iSCSI/452 : SUCCESS in 2h 14m 16s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f25d4caef7cddb609f32a5bb498c13e0443e192","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-19 06:36:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8b8422ebc9038321c07dea5ba31516d252375763","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-20 01:02:32.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781105-1/PVME-iSCSI : SUCCESS in 1:01:26\n* PVME-FC http://os-logs.tristero.net/dell-ci/781105-1/PVME-FC : SUCCESS in 54:05.76","accounts_in_message":[],"_revision_number":1},{"id":"7fd15f0d95cc4cc194a5465b97a43f58735d9490","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-20 04:20:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/781105/1/check/DellEMC_PowerFlex/7ab82f8/DellEMC_PowerFlex/609 : FAILURE in 2h 50m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"cb36c83d952b23fb2557b00b18de98480b7e018e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-20 04:55:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/781105/1/check/iscsi-driver-nimble/3795123/3795123.html : SUCCESS in 2h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"926d572bd89bf57776bea3a9b62c7aa6bc3b5027","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-20 08:35:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-1/ns5_nfs : SUCCESS in 5h 3m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-1/ns5_iscsi : SUCCESS in 5h 2m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ad53927599c0bdcec8b6ef62763b3b92258eacf8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-21 03:00:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/781105/1/check/DellEMC_Unity_FC/73f542d/DellEMC_Unity_FC/914 : SUCCESS in 1h 26m 06s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/781105/1/check/DellEMC_Unity_iSCSI/ff241dc/DellEMC_Unity_iSCSI/866 : SUCCESS in 1h 57m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8cfc32591f2483e211907c9035b206c8a529781d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-21 04:26:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d20c10207a75cf44907e68edc3f710ea50b3666a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-21 11:03:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/781105/1/check/DellEMC_XtremIO_FC/977df79/DellEMC_XtremIO_FC/1079 : SUCCESS in 2h 03m 30s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/781105/1/check/DellEMC_XtremIO_iSCSI/e16d890/DellEMC_XtremIO_iSCSI/1083 : SUCCESS in 1h 47m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"357ad015615e884b7f1cdaa33dec3694eb3780e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-21 12:46:57.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"fec10537e37b756a518baf6a6a5c81ac4fba715b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 01:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/781105/1/20210322073035/NEC-ISCSI : SUCCESS in 2h 16m 47s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/781105/1/20210322073035/NEC-FC : SUCCESS in 2h 30m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5890211acb0d7556bf17c72900358c63da3d82bf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-23 01:25:37.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nThis definitely makes the code easier to understand.","accounts_in_message":[],"_revision_number":1},{"id":"e486972f4d6f42b76d877ec0b72f71a85c5654ef","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-24 17:48:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good.  Thank you.","accounts_in_message":[],"_revision_number":1},{"id":"aed3052dc9ed96ed1dba5c5f8ef193d7fa9a55ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-26 11:26:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d47d54269d66065cd33cbc73b26c3a87062a6e5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-26 11:35:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"04771b3f90c2d089c5f80d7365fc4aac538486a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:18:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b93a4a1699d6470cbb8600a9d8d23365 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a4102ca04286212d4c747f74b036c251db79791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:53:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb15843c00044c3884ab53d302787ccc : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d01b8660034e9ab12340c240bdda4f : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf35f5a3159142f0a3c0e20f2cc5adaf : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88179d927dab4172873c9e0c0d6ca7b0 : SUCCESS in 11m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02f93004e91347c4bac8d7608f64a683 : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f205cfc726c14ca4a43b6cb013660e85 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c49fd9b48e0a40ea877e433fd61e314e : SUCCESS in 51m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36ce67906bf24ca98340ea53c87ef318 : SUCCESS in 1h 13m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/85c1508d3b5a45dd92ca5e1173a6e0b2 : SUCCESS in 15m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f64fdf019d0f4ea0aaf79f0d3c7334ce : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e708e1d6c6547089a66ec536e04539c : SUCCESS in 9m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/33e2a777878c418ea6e1ceb5a8652804 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89b08c3c7f2749d58c022daf5b7fe27d : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3fc4a18879804397ac85a9a536136918 : FAILURE in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e511f3a8f8b438e907b5222a49fbc97 : SUCCESS in 1h 24m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/17ab87c0c2a641089eccfafc4b32baf8 : FAILURE in 1h 24m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00623fd41ab44aff8f6cf1c3180d2d96 : SUCCESS in 1h 36m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/36bbb6a786354aafa402dba216625868 : SUCCESS in 1h 23m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5be6f6edb3914bbca82fb0e8fd3864c8 : FAILURE in 1h 28m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6498c074174b44d1b7703ce4c6505afe : SUCCESS in 1h 25m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4647146000864d478505f4c6893431bd : SUCCESS in 2h 11m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ffbfca16b094307b85f7d986220d4c4 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ab92382df4a77f682b33479d0c119d4b6799fa0d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-26 17:54:00.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nStill LGTM.  (Changes between PS1 and PS2 are from revisions of a different patch that modifies the same file.)","accounts_in_message":[],"_revision_number":2},{"id":"f38e3ccf40c2d3a0b8ae4820f63e0bd64928e767","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-26 19:33:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/781105/2/upstream-check/cinder-cDOT-iSCSI/64fd396/ : SUCCESS in 39m 14s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/781105/2/upstream-check/cinder-cDOT-NFS/5095e29/ : SUCCESS in 3h 08m 09s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/781105/2/upstream-check/cinder-cDOT-FCP/0932552/ : FAILURE in 2h 34m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"22a5f277ba81d81612a6eee4cde819a5689aabf6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-26 20:12:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/781105/2/upstream-solidfire-check/cinder-solidfire-iSCSI/041d7fe/ : FAILURE in 1h 32m 29s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"345a96d7b35c1c14f4e01e6e94169cc8a410ae06","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-26 21:40:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/05/781105/2/check/DellEMC_PowerStore_iSCSI/007f16d/DellEMC_PowerStore_iSCSI/726 : FAILURE in 38m 12s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/05/781105/2/check/DellEMC_PowerStore_FC/4ba8a3d/DellEMC_PowerStore_FC/720 : FAILURE in 36m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"8716d0f333bbbb8f5de860dbd641c8d33d0309a1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-27 00:42:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/781105/2/check/dsvm-tempest-cisco-zonemanager-job/c8b92cf/ : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":2},{"id":"fa135db0a27533061332c501fb8f914dd0d3c4eb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-27 06:33:52.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-2/ns5_nfs : SUCCESS in 5h 40m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-2/ns5_iscsi : SUCCESS in 5h 32m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"253dac74c811dd1d532ed977a79dab979775428b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-27 08:38:09.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7d52b38e86fef5cec6769153de6b43155294bde6","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-27 13:18:38.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781105-2/Lenovo-iSCSI : SUCCESS in 46:09.20\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781105-2/Lenovo-FC : SUCCESS in 43:02.30","accounts_in_message":[],"_revision_number":2},{"id":"d6f5742bec39301d80f75667a6dc96fa81f54cac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-27 16:10:28.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c8c33c0d080b4faf973dfe97880b69cf : POST_FAILURE in 1h 19m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/1b47bb79911f4be49eb0da38a79cbb59 : POST_FAILURE in 1h 15m 54s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud finger://zuul-v3-executor.novalocal/8e851f81959744b59b83f236eef2f093 : POST_FAILURE in 1h 46m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8d8e0a074a7aae8612eb90837cf6aac20a29069a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-27 18:53:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8196e5afcef7dab972841f762ef1f6fe0d35af8c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-27 22:15:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/781105/2/check/iscsi-driver-nimble/2297e29/2297e29.html : SUCCESS in 1h 51m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e9e2c58d09abc171cae6c95cd0ae854c44c34ba9","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-28 00:45:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/781105/2/check/DellEMC_SC_FC/63b89fa/DellEMC_SC_FC/525 : SUCCESS in 2h 00m 16s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/781105/2/check/DellEMC_SC_iSCSI/c488861/DellEMC_SC_iSCSI/522 : SUCCESS in 2h 08m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"b72d3a07a9fc2563a5be2dcfc3e61c11c9a89091","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-28 05:14:12.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"87f8401a71d79a9ff22555af1c2183cd2d2583d8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-28 06:43:47.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/781105/2/check/DellEMC_Unity_FC/f5adcc6/DellEMC_Unity_FC/1006 : SUCCESS in 1h 32m 36s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/781105/2/check/DellEMC_Unity_iSCSI/0dab64a/DellEMC_Unity_iSCSI/947 : SUCCESS in 1h 48m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"969beb6f330ab537194d164e25e92cf46d9a5b34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-28 09:19:26.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7c11915e36f546cafd01d1cc73fcb330c4bb11a7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-28 18:13:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/781105/2/check/DellEMC_XtremIO_FC/1de0cc3/DellEMC_XtremIO_FC/1161 : SUCCESS in 2h 03m 20s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/781105/2/check/DellEMC_XtremIO_iSCSI/972fc8d/DellEMC_XtremIO_iSCSI/1179 : SUCCESS in 1h 45m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"db6030554b0f88367280cee298a61a8e8f5d2e8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-28 20:32:45.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b428c74fde5bcd1e2e1db749096c5e3fd5159f86","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-29 00:02:40.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/781105/2/check/DellEMC_PowerFlex/5806bb6/DellEMC_PowerFlex/679 : ABORTED in 4h 00m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6841f79c179bf52805952dff3ec1582cb0b1bcae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-29 02:48:41.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7014f0158aff66fbab5363abe7c6fa6c73c34bbb","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-29 05:18:57.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781105-2/PVME-iSCSI : SUCCESS in 55:27.24\n* PVME-FC http://os-logs.tristero.net/dell-ci/781105-2/PVME-FC : SUCCESS in 55:52.62","accounts_in_message":[],"_revision_number":2},{"id":"edf5071bc00b0994595c170e778aed3e521ea5d4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-29 07:00:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/781105/2/check/DellEMC_VNX_iSCSI/6bfe0ce/DellEMC_VNX_iSCSI/660 : SUCCESS in 3h 12m 01s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/781105/2/check/DellEMC_VNX_FC/f7fb3e4/DellEMC_VNX_FC/706 : SUCCESS in 1h 30m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"a15ccbe8c5edf9ca86c88f76a9b3ce57d8314f49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-29 09:41:50.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"81be915067cc430fc86bb78b2d275747117f4013","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-29 22:08:55.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781105-2/Seagate-iSCSI : SUCCESS in 42:24.74\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781105-2/Seagate-FC : SUCCESS in 38:22.05","accounts_in_message":[],"_revision_number":2},{"id":"365f1e6e4b091f5ada18146bf064e3f032704a48","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-30 00:48:05.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781105-2/HPMSA-iSCSI : SUCCESS in 58:20.40\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781105-2/HPMSA-FC : SUCCESS in 33:49.03","accounts_in_message":[],"_revision_number":2},{"id":"b03e3ad043c94c73381f5f6842e66c0c1a4e15c0","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 03:50:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/781105/2/20210330103527/NEC-FC : SUCCESS in 2h 13m 59s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/781105/2/20210330103527/NEC-ISCSI : SUCCESS in 2h 14m 40s","accounts_in_message":[],"_revision_number":2},{"id":"9ba855aadf373030b8b1d0ed5808477031fad42e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-30 13:35:00.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-2/ns5_nfs : SUCCESS in 5h 22m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-2/ns5_iscsi : SUCCESS in 5h 18m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"39f9c78b6126bc6e5b298fc8eb7c2ad3741a1b12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-30 13:48:21.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c1130588a5d7d0061599c596c677ee468ddfc76f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-30 14:23:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3669adcef066b306b8d435af41cc5ee6b8a6d417","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 18:57:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/781105/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/59e948d/59e948d.html : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":4},{"id":"9ad28eae3d4a65cbaa4968b81092f5a81845f89a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 19:28:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/781105/4/check/iscsi-driver-primera/48eebb9/48eebb9.html : FAILURE in 1h 02m 22s","accounts_in_message":[],"_revision_number":4},{"id":"d55cc40d35e0fcf051dacadaa3a537e80cdded2c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-30 20:16:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/781105/4/check/dsvm-tempest-cisco-zonemanager-job/85d17a5/ : SUCCESS in 1h 22m 04s","accounts_in_message":[],"_revision_number":4},{"id":"134b8019ff1a424d2dff1b3214fef2ada997f26e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 23:48:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/781105/4/check/iscsi-driver-nimble/610755c/610755c.html : FAILURE in 1h 17m 37s","accounts_in_message":[],"_revision_number":4},{"id":"544769465bd51300c6ac5900e1b10dd79a1e9671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:11:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1752d1b2c95344509ef3c5b2430cb5bf : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fcc7cc7967346fbbb315940bc6b3f29 : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/179a3a5d410e4bf1848704592e1c7125 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a51a9c73e1e94eaf904417480da03baa : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/001997cc290741758b9adbc497cfb175 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fae1ccfe901450fbca2e14e0b032c92 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/31a700c5aceb42afbed781852ee6b04d : SUCCESS in 57m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/999839d27aef499087e58c7ac0d84332 : SUCCESS in 1h 14m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a51f7c82807460eaf89a91593231a42 : SUCCESS in 13m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76f72748fb2c47aab5e30ca44c2cf963 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/006b15b4022242e7829d4228f7b392d1 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa408231a82547018e7c369ccaa59d9f : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b82a5a3718464b559c859116f53882fc : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a6cd2630d094f26922c233c78071507 : FAILURE in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0bb335c580c4a72891a8c3de9a5bcd6 : SUCCESS in 1h 20m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c1623a2061445f1b2f3134d30af3a15 : FAILURE in 1h 28m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/804409a637b34b9fa34e3f367d7fdf3e : SUCCESS in 1h 40m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/58fa5e7f7e7543f2ab99774569ff4a0d : SUCCESS in 1h 19m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad4c95d79a9745d7ad6475dd7864445a : SUCCESS in 1h 20m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89430974c34a4245a9a43dc135fae0ae : SUCCESS in 1h 20m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04d601c65fe54dea8c43e49ecd16639a : SUCCESS in 2h 02m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb43479a4e754186b2fe6f868a7c2731 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":4},{"id":"ecdf7d93398c9d948c7acaf916b951f47ab8735b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-31 01:42:35.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/781105/4/upstream-solidfire-check/cinder-solidfire-iSCSI/8911d56/ : FAILURE in 1h 28m 33s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"b7d907065c83eea09bc3a2b9935ffdd044341dda","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-31 04:17:47.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781105-4/Lenovo-iSCSI : SUCCESS in 46:22.71\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781105-4/Lenovo-FC : FAILURE in 45:48.76","accounts_in_message":[],"_revision_number":4},{"id":"8ebe9bfc82745ff681076cb2f49ae66b8e654d5d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-31 04:37:41.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_SC_FC/9d23b5d/DellEMC_SC_FC/548 : SUCCESS in 2h 10m 12s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_SC_iSCSI/80b1b3e/DellEMC_SC_iSCSI/545 : SUCCESS in 2h 05m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"24940ce7639e101a6393723183062d4e3ea321ce","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-31 04:50:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/781105/4/upstream-check/cinder-cDOT-iSCSI/5ea090f/ : SUCCESS in 41m 17s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/781105/4/upstream-check/cinder-cDOT-NFS/fdd22dd/ : SUCCESS in 1h 56m 19s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/781105/4/upstream-check/cinder-cDOT-FCP/40fe5a6/ : SUCCESS in 1h 08m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"cac0235fa10cb005a4c8098c024d80cd81d7b536","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-31 06:40:11.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-4/ns5_nfs : SUCCESS in 6h 1m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-781105-4/ns5_iscsi : SUCCESS in 5h 30m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"7744b788745600f7fe905cd5c9dd7a418ded6240","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 07:05:42.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/781105/4/check/DellEMC_PowerFlex/e0873d0/DellEMC_PowerFlex/698 : FAILURE in 3h 04m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"527a0e03f8c805f82506bdb6fb7889d5dca92dd3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 07:33:46.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_XtremIO_FC/fa3d4c7/DellEMC_XtremIO_FC/1192 : SUCCESS in 2h 03m 58s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_XtremIO_iSCSI/a49de98/DellEMC_XtremIO_iSCSI/1210 : SUCCESS in 1h 45m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"9c1a36150408ec0d4651933ef7b69a13189edd8a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-31 07:48:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_Unity_FC/b5ac21a/DellEMC_Unity_FC/1037 : SUCCESS in 1h 35m 35s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_Unity_iSCSI/e19b8e0/DellEMC_Unity_iSCSI/978 : SUCCESS in 1h 58m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"0a482f56d702edbce70402157b08e0607aae66f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-31 11:43:16.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"a16f44112c45e6342d61d5b5cfcfeeb94d1f5b9a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-31 14:33:28.000000000","message":"Patch Set 4: Code-Review+2\n\nThis still looks ok to me.","accounts_in_message":[],"_revision_number":4},{"id":"b93b04bea4a89eaca798a7b92790e1c7402597fb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-31 14:50:01.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/781105/4/check/tempest-dsvm-full-focal-py3/240baf8/ : SUCCESS in 2h 11m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/781105/4/check/devstack-plugin-ceph-dsvm-focal-py3/71ea353/ : SUCCESS in 2h 29m 21s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/781105/4/check/devstack-plugin-ceph-dsvm-focal-py3-newcloud/2aa5a4a/ : TIMED_OUT in 3h 15m 01s","accounts_in_message":[],"_revision_number":4},{"id":"e481da4e4f9d626a721c4c6d179193d6dda93f36","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-31 16:56:37.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781105-4/Seagate-iSCSI : SUCCESS in 43:45.67\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781105-4/Seagate-FC : SUCCESS in 39:38.48","accounts_in_message":[],"_revision_number":4},{"id":"81e36182b90c8d705af7e6fcfcd91244901a707e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-31 16:58:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fd76acc064f73314826800b468fa6a1c895a3616","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-31 18:45:22.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"f06c4326fcd4664eabc1576f4ebfc78f7b85d407","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-03-31 20:00:05.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/781105/4/fc : SUCCESS in 0h 48m 34s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/781105/4/iscsi : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":4},{"id":"2aee7827a05b497275937d45c8884744b902e148","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-01 01:53:08.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781105-4/HPMSA-iSCSI : FAILURE in 1:00:17\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781105-4/HPMSA-FC : SUCCESS in 33:55.10","accounts_in_message":[],"_revision_number":4},{"id":"02f43054ae4540f1e908ee082c604ca814d02c0d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-01 02:52:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_VNX_iSCSI/9d69d57/DellEMC_VNX_iSCSI/680 : SUCCESS in 3h 14m 42s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_VNX_FC/5467754/DellEMC_VNX_FC/729 : SUCCESS in 1h 33m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"960d9ecdc1b5d4908fc41bda9ea3930a3574a222","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-01 09:30:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"4a737b62de9afdaa0b4bc31b179d11ba2ca09d24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:21:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8705ef8f78cf6fdf979c4cbb692a46c2cf62b3be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:38:58.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a15426d12566e04c58dea536529433fc578510bd","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-01 16:40:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/781105/4/20210401230929/NEC-ISCSI : SUCCESS in 2h 13m 14s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/781105/4/20210401230929/NEC-FC : SUCCESS in 2h 31m 24s","accounts_in_message":[],"_revision_number":4},{"id":"98c206c9b4e3b44feb98ff3fb397bba5540be615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:00:28.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"cf5cc3a3ad0646b05967917d165f3517d73213b2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-02 03:36:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"eb9163fe127c2144bb95e82b466bca71bc9404bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 04:07:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7fffd790be147748d0cdd17d8623650 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e69062d4586f4d34935fa865f045a1e7 : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"658089eb34a458f7f4995ee4f4c261694ec90338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 06:06:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4cbf94a8a3d4773afcfd769c62403ad : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37106df42af418e88672435e53dc512 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cbac5df28324eafb6730f5176d1ce4b : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/623216277179461499b9916992c6bf50 : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c5df9f922ef4dfca646fbc6f83e03a0 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d5a501ace74292a94becb48ced0634 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/b9df6524b8bf486e8d246524af148927 : SUCCESS in 1h 14m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/214c3b604e4e421fb9a2b7841ca745e5 : SUCCESS in 1h 23m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b4a986b832c42399bf66c11fc87ef55 : SUCCESS in 13m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8291228147f4c60ac2120be90e3e92b : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5abb3deef724d58afb3194f2e345289 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/360cfb7eab444ca0a08bf5683d0724e3 : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/41467057fa7d4efeae589efb7361dc51 : SUCCESS in 47m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c198997605a94a389be8ed981a227f53 : FAILURE in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dc39f9f2ef104709abd6abffede78941 : SUCCESS in 1h 13m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c850171d12b3483294755782776b5dc3 : FAILURE in 1h 32m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5da12be124be4861b63404a8d866e420 : SUCCESS in 1h 48m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc372063728a493eb57877c8dc59f006 : SUCCESS in 1h 19m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29f35d41945147b5a53b898a98b94d42 : SUCCESS in 1h 19m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ecc9f0e971f4c4089fe3419ef100ae7 : SUCCESS in 1h 25m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86d0fcd42c6045889c4310e0f44f76fa : SUCCESS in 2h 24m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f9f539d7be4408a818d366343f28efe : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e6c1e150fc7263f3bcde0be17d20b6b96493fd32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 06:07:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"340ddd7d3fd155ed03efa0cbe4adf24262bd8ca2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-02 06:14:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/781105/4/check/dsvm-tempest-cisco-zonemanager-job/d418ddf/ : SUCCESS in 1h 21m 49s","accounts_in_message":[],"_revision_number":4},{"id":"32e54c6deb1046202a0324682c610be8ab3280d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-02 07:48:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/781105/4/upstream-check/cinder-cDOT-iSCSI/6bc6cae/ : FAILURE in 40m 46s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/781105/4/upstream-check/cinder-cDOT-NFS/99811f6/ : SUCCESS in 40m 24s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/781105/4/upstream-check/cinder-cDOT-FCP/59fcd10/ : SUCCESS in 1h 41m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"94387d0ac080bfa799318b50fd9a37a45301602e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 08:08:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/584da5f283814eb498ef478df97ec8ad : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b781ab9f47cf47b4b77259dd162171bb : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf0f1d614c14e24a5eff041004a6443 : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a83a9c7b8c7d4a968cee4379f12699a1 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5367b60e26f49e8a6623871529d5661 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/79a29dd0f80148cd8c972925e6d48e16 : SUCCESS in 1h 07m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f733349793bc4c378fdb1b32f9b2474b : SUCCESS in 1h 25m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bdd2298fa8aa44a38583f94c1b912857 : SUCCESS in 1h 16m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63d0688391c9462f9469dd4fc0f17ca4 : SUCCESS in 1h 27m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a925e72839643b280b038f432e3b531 : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":4},{"id":"5ee38a3f7f4fac836ec7ed47a2c750b18227f6a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 08:08:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8bb02e724ade317820c883722f8e56283eebc813","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-02 09:59:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/781105/4/upstream-solidfire-check/cinder-solidfire-iSCSI/7e2fabd/ : FAILURE in 1h 32m 22s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"17e82d6b2f1cee4846c96831bedc3c92a9a3e0ab","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-02 10:12:03.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/781105/4/check/tempest-dsvm-full-focal-py3/ced744f/ : SUCCESS in 1h 51m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/781105/4/check/devstack-plugin-ceph-dsvm-focal-py3/2de5021/ : SUCCESS in 2h 12m 51s","accounts_in_message":[],"_revision_number":4},{"id":"a2d154c77251b63e5be055b2db2d0703da3500cb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-02 19:54:09.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_SC_FC/bad0896/DellEMC_SC_FC/574 : SUCCESS in 1h 58m 50s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_SC_iSCSI/ff602e3/DellEMC_SC_iSCSI/571 : SUCCESS in 2h 27m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"a25f34b15416e147c56dcedd6fd71a67d01d62f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-02 22:39:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"8451aa1b9f804b02428cf985a763f4729c4ff05f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 02:42:56.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_XtremIO_FC/0d4e163/DellEMC_XtremIO_FC/1225 : SUCCESS in 2h 03m 48s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_XtremIO_iSCSI/a0b45b6/DellEMC_XtremIO_iSCSI/1244 : SUCCESS in 1h 44m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"42e75df92ab360305f57c736821fe87c0713b78e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-03 03:01:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_Unity_FC/896fed3/DellEMC_Unity_FC/1073 : FAILURE in 1h 57m 06s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_Unity_iSCSI/e05bb93/DellEMC_Unity_iSCSI/1011 : FAILURE in 2h 07m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"13d471dcf6ffac0a24fadc37646fb667dbb8e245","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-03 03:51:28.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781105-4/PVME-iSCSI : SUCCESS in 59:32.95\n* PVME-FC http://os-logs.tristero.net/dell-ci/781105-4/PVME-FC : SUCCESS in 56:38.67","accounts_in_message":[],"_revision_number":4},{"id":"a30d139becf1e8ab567c7d17bbf1714c94ce5123","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 04:09:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/781105/4/check/DellEMC_VNX_iSCSI/f2dce92/DellEMC_VNX_iSCSI/702 : FAILURE in 1h 12m 38s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/781105/4/check/DellEMC_VNX_FC/43b28eb/DellEMC_VNX_FC/751 : SUCCESS in 2h 20m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"71ee37ece0d82552beefa6c2c0ab0b9a8079b598","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 05:38:00.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"cc635e43d23a011ea7640657c55a452a3db8fab6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 09:00:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/781105/4/check/DellEMC_PowerFlex/ca2af88/DellEMC_PowerFlex/723 : SUCCESS in 2h 42m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"a140cf25fcab6cf66c1b3392639409212abea3c5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 10:23:43.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/781105/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"70e24bb48d14909d4e1a78b41e75f7d0a40d6e2f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-04-04 14:37:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/781105/4/ : FAILURE in 1h 54m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c7eaa54c19ea0a772ff9d3cfcb659518355519e5","author":{"_account_id":31305,"name":"zhubingyang","email":"yang77498441@126.com"},"date":"2021-04-04 15:23:48.000000000","message":"Patch Set 4:\n\nrun INSPUR-CI","accounts_in_message":[],"_revision_number":4},{"id":"9c5b194304f6cdb5598d51ca2239d755031d0cb3","author":{"_account_id":31305,"name":"zhubingyang","email":"yang77498441@126.com"},"date":"2021-04-05 00:21:37.000000000","message":"Patch Set 4:\n\nrun-INSPUR CI","accounts_in_message":[],"_revision_number":4},{"id":"33964c986fa8d7ccdb46844e89e0b6ec52967c65","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-06 04:36:46.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781105-4/Lenovo-iSCSI : SUCCESS in 46:19.24\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781105-4/Lenovo-FC : SUCCESS in 43:49.57","accounts_in_message":[],"_revision_number":4},{"id":"55e484ea2edafe64c5707d1fada3f9ad094836a3","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-06 05:50:06.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781105-4/Seagate-iSCSI : SUCCESS in 44:11.87\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781105-4/Seagate-FC : SUCCESS in 37:35.05","accounts_in_message":[],"_revision_number":4},{"id":"20d68b8a3266982167aae2ed5879472821416b83","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-06 17:51:42.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781105-4/HPMSA-iSCSI : FAILURE in 1:06:31\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781105-4/HPMSA-FC : SUCCESS in 34:33.87","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cdd60605f45f9ed624150d1310d8bce2cbef48c3","revisions":{"26bfba09845248e3223ffa4bae08b6a31db13f9c":{"kind":"REWORK","_number":1,"created":"2021-03-17 14:00:25.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/05/781105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/781105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/781105/1"}}},"commit":{"parents":[{"commit":"7a1e3fc15403e5416635632c7053a473b10ef1b8","subject":"Fix quota usage duplicate entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7a1e3fc15403e5416635632c7053a473b10ef1b8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:58:56.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-17 09:02:00.000000000","tz":60},"subject":"Update cinder manage quota commands","message":"Update cinder manage quota commands\n\nThis patch updates the cinder manage quota commands to leverage the new\nmethod _get_sync_updates from cinder db, allowing us to simplify our\ncode.\n\nChange-Id: I099d101a3f2bcfbd2026306081a9858c610ab1ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26bfba09845248e3223ffa4bae08b6a31db13f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26bfba09845248e3223ffa4bae08b6a31db13f9c"}]},"branch":"refs/heads/master"},"a8d5a8ff7bb6539bd528d3d9619a290663f7ba4e":{"kind":"REWORK","_number":2,"created":"2021-03-26 11:26:48.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/05/781105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/781105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/781105/2"}}},"commit":{"parents":[{"commit":"6453d14a44192170b98a6ef65623da26252159d6","subject":"Fix quota usage duplicate entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6453d14a44192170b98a6ef65623da26252159d6"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:58:56.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-26 11:26:15.000000000","tz":60},"subject":"Update cinder manage quota commands","message":"Update cinder manage quota commands\n\nThis patch updates the cinder manage quota commands to leverage the new\nmethod _get_sync_updates from cinder db, allowing us to simplify our\ncode.\n\nChange-Id: I099d101a3f2bcfbd2026306081a9858c610ab1ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8d5a8ff7bb6539bd528d3d9619a290663f7ba4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8d5a8ff7bb6539bd528d3d9619a290663f7ba4e"}]},"branch":"refs/heads/master"},"8f78194143ec7d955b09638b1942866cfdaf3b22":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-30 13:48:21.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/05/781105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/781105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/781105/3"}}},"commit":{"parents":[{"commit":"59fd7c8697884ca60fb89ab1ff0e64d59019ef71","subject":"Fix quota usage duplicate entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59fd7c8697884ca60fb89ab1ff0e64d59019ef71"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:58:56.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-30 13:46:39.000000000","tz":120},"subject":"Update cinder manage quota commands","message":"Update cinder manage quota commands\n\nThis patch updates the cinder manage quota commands to leverage the new\nmethod _get_sync_updates from cinder db, allowing us to simplify our\ncode.\n\nChange-Id: I099d101a3f2bcfbd2026306081a9858c610ab1ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f78194143ec7d955b09638b1942866cfdaf3b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f78194143ec7d955b09638b1942866cfdaf3b22"}]},"branch":"refs/heads/master"},"cdd60605f45f9ed624150d1310d8bce2cbef48c3":{"kind":"REWORK","_number":4,"created":"2021-03-30 14:23:56.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/05/781105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/781105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/781105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/781105/4"}}},"commit":{"parents":[{"commit":"1fb0767d8874f993e4ac6d9fda92933f1c5b6d3f","subject":"Fix quota usage duplicate entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1fb0767d8874f993e4ac6d9fda92933f1c5b6d3f"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:58:56.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-30 14:23:08.000000000","tz":120},"subject":"Update cinder manage quota commands","message":"Update cinder manage quota commands\n\nThis patch updates the cinder manage quota commands to leverage the new\nmethod _get_sync_updates from cinder db, allowing us to simplify our\ncode.\n\nChange-Id: I099d101a3f2bcfbd2026306081a9858c610ab1ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdd60605f45f9ed624150d1310d8bce2cbef48c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdd60605f45f9ed624150d1310d8bce2cbef48c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
