)]}'
{"id":"openstack%2Fcinder~781106","triplet_id":"openstack%2Fcinder~master~I487157d7092fa4cee1cd72cf01ce7031c94f553f","project":"openstack/cinder","branch":"master","topic":"unused-quota-classes","hashtags":[],"change_id":"I487157d7092fa4cee1cd72cf01ce7031c94f553f","subject":"Prevent quota and reservations to go into negative","status":"MERGED","created":"2021-03-17 14:00:25.000000000","updated":"2021-04-06 11:33:18.000000000","submitted":"2021-04-02 08:09:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781106-unused-quota-classes","meta_rev_id":"6b46e18e21c22a470c345be7cb9f1efd89f001ba","_number":781106,"virtual_id_number":781106,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-02 08:09:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-04-06 11:33:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-31 14:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-04-01 13:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-04-06 11:33:18.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 16:16:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 17:33:50.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-17 20:10:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 20:12:02.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-17 20:16:26.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 21:50:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 00:36:21.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-18 07:23:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-18 14:14:16.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-18 15:07:24.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-19 12:20:17.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-19 14:39:42.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 09:42:09.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 04:56:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-21 13:07:44.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 03:23:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 02:20:41.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-24 17:46:45.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-27 03:16:06.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-28 15:08:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-29 12:28:07.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-29 14:06:29.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-31 18:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-04-03 14:33:40.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ed7be5075c3743fc90196790d660503771bbdb94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-17 14:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87cc8cf3c46b78f3405d531078370727d99df1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:16:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c32b97bc10fe4963b57a7d99cc0dcc4b : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013d13112b0b4225b884c79bcdb81555 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddca9f6ab5ba41d8922e409d562a6467 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6cc3085bd5b4fcb8c9acef56fa14a9f : SUCCESS in 14m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66ecc1f19a454ae5bb1df5bdf839efbd : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08466e1a8a2c45b59e7557ea250d9dc4 : SUCCESS in 18m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/787e03dc691044b9a233aae86402d5e0 : SUCCESS in 1h 23m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/554a6e1467d24e8bab0e6cf1026c19b9 : SUCCESS in 1h 11m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/74d928de4b5e4270b7ea7725f2615b2c : SUCCESS in 15m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1cff70f3a9241d78987ac235a681851 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4610f42f8bc546d6b9971a3415486676 : SUCCESS in 9m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5356b0da14314ec3b967ade3edfafb3f : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8f54aa26283c42cc8da0a884b921d358 : SUCCESS in 40m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b3354f152354fce876eb435a49b1aca : FAILURE in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b071cfc4ee8448129306a806f7907284 : SUCCESS in 1h 26m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9bf5aefaba2b4c96803ece45467957a1 : FAILURE in 1h 41m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/87cfbffc2d654bb4a184506330594624 : SUCCESS in 1h 58m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d714b7493e09456dbfe0c6928e81c599 : SUCCESS in 1h 18m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1091925dc11e4be48b4d60346f51a177 : SUCCESS in 1h 38m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ccfbaa44dae48988d4c97d7703a6945 : FAILURE in 1h 39m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a9d18bbedf646eba5158f19193b0f8c : SUCCESS in 2h 01m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a746ae6a497e49818ae9a836656766f1 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ce4f9976b766fdd7eb39d0df033e72d3bb910e69","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-17 17:33:50.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/06/781106/1/check/dsvm-tempest-TOYOU-ACS5000-driver/9636725 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"eaba8cbaec46c587e20af5204f5385769a9841cb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-17 20:10:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/06/781106/1/upstream-check/cinder-cDOT-iSCSI/51ff039/ : FAILURE in 4m 02s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/06/781106/1/upstream-check/cinder-cDOT-iSCSI-IPv6/19f7104/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/06/781106/1/upstream-check/cinder-cDOT-NFS/5a2b22b/ : FAILURE in 3h 50m 18s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/06/781106/1/upstream-check/cinder-cDOT-NFS-IPv6/0dcbae1/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/06/781106/1/upstream-check/cinder-cDOT-FCP/ae9240c/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6e5ccc540889caf49912b349cf5e9b4ba56a2fa","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-17 20:12:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/06/781106/1/upstream-solidfire-check/cinder-solidfire-iSCSI/12f76eb/ : FAILURE in 3h 50m 05s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"0611806a6321e540c2945de237e4fe3dfb9cf57f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-17 20:16:26.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/781106/1/check/tempest-dsvm-full-focal-py3/e4f73f7/ : SUCCESS in 2h 03m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/781106/1/check/devstack-plugin-ceph-dsvm-focal-py3/7fa8301/ : SUCCESS in 2h 07m 03s","accounts_in_message":[],"_revision_number":1},{"id":"06586b35fcc3e7e99e90c6169d50e4d7eba36926","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-17 21:50:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/06/781106/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ae0d3d5/ae0d3d5.html : FAILURE in 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e06833a8c11144329cf7d5ccb43fa720340eced9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-18 00:36:21.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/06/781106/1/check/DellEMC_PowerStore_iSCSI/1f95b33/DellEMC_PowerStore_iSCSI/635 : SUCCESS in 56m 53s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/06/781106/1/check/DellEMC_PowerStore_FC/0654070/DellEMC_PowerStore_FC/629 : SUCCESS in 53m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"c9214b01b0da5f67db9e1047c5f4475249bd32d5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-18 07:23:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/06/781106/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2240eb1/ : POST_FAILURE in 2h 19m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/06/781106/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/395ac47/ : SUCCESS in 2h 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2348c3f3a90acd4fc9fc122d06a72e200014a037","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-18 14:14:16.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/06/781106/1/check/DellEMC_SC_FC/8b19d3a/DellEMC_SC_FC/450 : SUCCESS in 1h 56m 46s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/06/781106/1/check/DellEMC_SC_iSCSI/dd39c64/DellEMC_SC_iSCSI/447 : SUCCESS in 2h 32m 06s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"c43ef5f245d7a1a36c800db645869dbae81c927b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-18 15:07:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/781106/1/check/dsvm-tempest-cisco-zonemanager-job/8d64e2a/ : FAILURE in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"803e7df76453c51491783c52d8047d70ddfd434f","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-19 12:20:17.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781106-1/PVME-iSCSI : SUCCESS in 1:01:19\n* PVME-FC http://os-logs.tristero.net/dell-ci/781106-1/PVME-FC : SUCCESS in 56:23.82","accounts_in_message":[],"_revision_number":1},{"id":"c41999ff32cf0d4831d88a29d7094e5bf8f71e44","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-19 14:39:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/06/781106/1/check/DellEMC_PowerFlex/3c0a3e0/DellEMC_PowerFlex/604 : FAILURE in 2h 50m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"63fe36fb99bb8a4a950a1c804f5bd42a5ff8f38f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-20 05:22:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/06/781106/1/check/iscsi-driver-nimble/5f886a5/5f886a5.html : FAILURE in 22m 16s","accounts_in_message":[],"_revision_number":1},{"id":"860a73de7d0873a71176bdce6f6340105e2f278f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-20 09:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-1/ns5_nfs : SUCCESS in 5h 10m 7s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-1/ns5_iscsi : SUCCESS in 4h 57m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4e21ab4a9a8f081ba691c572f48ff96ac34ba77b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-21 04:56:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/06/781106/1/check/DellEMC_Unity_FC/ddfb95d/DellEMC_Unity_FC/915 : SUCCESS in 1h 24m 33s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/06/781106/1/check/DellEMC_Unity_iSCSI/329c4eb/DellEMC_Unity_iSCSI/867 : SUCCESS in 1h 56m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"dbfff91099155ce0ac7433c37cd7746b9aec75c5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-21 13:07:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/06/781106/1/check/DellEMC_XtremIO_FC/8a838a1/DellEMC_XtremIO_FC/1080 : FAILURE in 2h 04m 15s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/06/781106/1/check/DellEMC_XtremIO_iSCSI/017425d/DellEMC_XtremIO_iSCSI/1084 : SUCCESS in 1h 46m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"3d22884d5decb6ef7b99015ce1d472444fd7a150","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 03:23:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/06/781106/1/20210322100104/NEC-FC : SUCCESS in 2h 19m 08s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/06/781106/1/20210322100104/NEC-ISCSI : SUCCESS in 2h 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0433d9313bbd0b09cae50500b50792b88f659493","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-23 02:20:41.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nThe commented tests are really helpful for understanding how this should work.","accounts_in_message":[],"_revision_number":1},{"id":"c2dcd408d4121074793fb41269432583073656f7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-24 17:46:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThere is a typo here, but in the interest of getting this in I think we can merge it and do a follow-up patch.","accounts_in_message":[],"_revision_number":1},{"id":"86f19129bc17b4e05d7ec74468c0172788594432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-26 11:26:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9709754ce037db4e426f8708915eca155ec0c87b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-26 11:28:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22729e2507dcc8ea5078edcaba12e8fd24de38a7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-26 11:35:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"cbb1a5fe110219ab0817afed8dbaf89cac109261","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:18:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a5fec8fcbd3443eadd3dcbc7a2987a5 : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e24c9e88eeb1c181b602d24c3ff5f98ceab83b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:39:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6518417fd9d457682f89d6a88b0bd9e : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263e382c0f9d4e6190e415dcbcc45782 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2febe5d25e8f4313addb0b1c1c2ee9bc : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/335a453c72854c5ca21fb03ef88b32e4 : SUCCESS in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37f8761d360b419fa24b9499072cc788 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0ac23f229942aa8413588e626ca3ad : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/264d8ddaa6b34d55a476e35bd7a0e76e : SUCCESS in 1h 06m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c67384326fe44bd9a80a57455224661 : SUCCESS in 1h 10m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6375753f13a04eeeb6dd8282d8eb8200 : SUCCESS in 12m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f781959230e42b794a8ac4c2d277413 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e811f5b8474400eba745699c7d47909 : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d844ad8e940040f7aa16dd28f1a95bf4 : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b66ab194d094dffa7fd68364d3e4f9a : SUCCESS in 45m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f2994d6b72a4da29f3d40622101eb73 : FAILURE in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/997885c90fb24af788c376191837eadd : SUCCESS in 1h 17m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7d77c0e1a19b44f88cd3b0a06a425252 : FAILURE in 1h 32m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/380c11f544c440a987d6bc258b283dc5 : SUCCESS in 1h 43m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3f270be1e52d4cd6a18fa70f871159aa : SUCCESS in 1h 06m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8c3f2a2ce4e43f79ede06b87604c8a1 : SUCCESS in 1h 43m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ca2362bf12af4c61b4774e5def29bfd9 : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/594015fa49cf445bb3119cf072527d51 : SUCCESS in 1h 59m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a53157e552d4fdf9b7d2937459ea990 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":2},{"id":"acb1ae21b5f790444e98ae5145ee8f1e48037f75","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-26 17:34:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/06/781106/2/upstream-solidfire-check/cinder-solidfire-iSCSI/4556228/ : FAILURE in 2h 47m 56s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"c982801dc01e8a24242032a2b3a665c559bd79e0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-26 17:57:42.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nThe revision looks good to me.","accounts_in_message":[],"_revision_number":2},{"id":"e489fdcea6db8239c0869e4cf28715e7b8586829","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-26 18:28:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/06/781106/2/check/DellEMC_PowerStore_iSCSI/dc9d3cb/DellEMC_PowerStore_iSCSI/721 : FAILURE in 38m 24s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/06/781106/2/check/DellEMC_PowerStore_FC/ebc321c/DellEMC_PowerStore_FC/715 : FAILURE in 36m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"8d6214975171cf32fdbbbad9da2094ce4a4535e2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-26 19:57:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/06/781106/2/upstream-check/cinder-cDOT-iSCSI/0610af4/ : FAILURE in 1h 00m 27s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/06/781106/2/upstream-check/cinder-cDOT-NFS/d736d27/ : SUCCESS in 43m 12s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/06/781106/2/upstream-check/cinder-cDOT-FCP/c8f4960/ : SUCCESS in 1h 16m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7df7b2d2462a2f2661350ab8981e9431235e7e8c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-27 01:26:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/781106/2/check/dsvm-tempest-cisco-zonemanager-job/7ae5c76/ : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2ab0aa8a80e3966f4df6f723074b667e376dfbd3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-27 03:16:06.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781106-2/Lenovo-iSCSI : SUCCESS in 47:51.82\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781106-2/Lenovo-FC : SUCCESS in 43:28.41","accounts_in_message":[],"_revision_number":2},{"id":"7da7cc6135b768d347733bc20e8c150ae1127884","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-27 09:17:27.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-2/ns5_nfs : SUCCESS in 5h 48m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-2/ns5_iscsi : SUCCESS in 5h 29m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"7c11efaaf8af54119e0024b119ca21a5cffd0900","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-27 13:56:28.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/06/781106/2/check/DellEMC_SC_FC/fe90cb6/DellEMC_SC_FC/520 : SUCCESS in 2h 06m 22s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/06/781106/2/check/DellEMC_SC_iSCSI/ba2c940/DellEMC_SC_iSCSI/517 : SUCCESS in 2h 10m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"2b76b85df63def264ee6354fa1ad8d6dea6a30a7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-27 21:26:41.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5b6f48b31865441293fab3a443e17454 : POST_FAILURE in 1h 16m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/a82140544aab4e20becfb87b881b8005 : POST_FAILURE in 1h 21m 14s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud finger://zuul-v3-executor.novalocal/b51accb1ccc345558997cfbf93697ca4 : POST_FAILURE in 1h 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1720ed98e1ec18dd0ba3ef81acd0941cae7a2df5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-27 22:49:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/06/781106/2/check/iscsi-driver-nimble/db55a2f/db55a2f.html : FAILURE in 29m 23s","accounts_in_message":[],"_revision_number":2},{"id":"04f30756293cb841f3d73f438ed109895bda0a00","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-28 04:01:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/06/781106/2/check/DellEMC_PowerFlex/53fa44a/DellEMC_PowerFlex/674 : ABORTED in 4h 00m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f1a3deff8154908a6ae1b9b14a18f53b1aa66b54","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-28 08:36:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/06/781106/2/check/DellEMC_Unity_FC/4e6d180/DellEMC_Unity_FC/1007 : SUCCESS in 1h 40m 40s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/06/781106/2/check/DellEMC_Unity_iSCSI/1956572/DellEMC_Unity_iSCSI/948 : SUCCESS in 1h 52m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"20065234329942948538ca3eef9b586287be588b","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-28 14:46:16.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781106-2/PVME-iSCSI : FAILURE in 1:42:09\n* PVME-FC http://os-logs.tristero.net/dell-ci/781106-2/PVME-FC : SUCCESS in 56:32.80","accounts_in_message":[],"_revision_number":2},{"id":"419985b5a4c6bd1c3f94f590b035922c8c257303","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-28 15:08:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/06/781106/2/check/DellEMC_VNX_iSCSI/d183ba2/DellEMC_VNX_iSCSI/655 : SUCCESS in 3h 12m 03s\n- DellEMC_VNX_FC http://publiclogs.emc.com/06/781106/2/check/DellEMC_VNX_FC/e941714/DellEMC_VNX_FC/701 : SUCCESS in 1h 21m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"28c8f4d9456296c9cdeaf50096ee7aec7a582fd0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-28 20:19:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/06/781106/2/check/DellEMC_XtremIO_FC/4404419/DellEMC_XtremIO_FC/1162 : FAILURE in 2h 06m 37s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/06/781106/2/check/DellEMC_XtremIO_iSCSI/247906e/DellEMC_XtremIO_iSCSI/1180 : SUCCESS in 1h 43m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"c776167a02100f15e6778100943aebba2bf901ba","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-29 12:28:07.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781106-2/Seagate-iSCSI : SUCCESS in 44:24.82\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781106-2/Seagate-FC : SUCCESS in 37:34.82","accounts_in_message":[],"_revision_number":2},{"id":"ba6ddbc7cfd1b864b9bb93f5f5c814132413f00c","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-29 14:06:29.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781106-2/HPMSA-iSCSI : FAILURE in 1:02:31\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781106-2/HPMSA-FC : SUCCESS in 34:42.65","accounts_in_message":[],"_revision_number":2},{"id":"6612e6af625340ceb777d4efd6c31388d1d6b454","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 06:05:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/06/781106/2/20210330125014/NEC-ISCSI : SUCCESS in 2h 10m 39s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/06/781106/2/20210330125014/NEC-FC : SUCCESS in 2h 15m 27s","accounts_in_message":[],"_revision_number":2},{"id":"63e7a1df9d3763ea4af71c9700dd3d7a617d918a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-30 13:48:21.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c21ea54c06b94cc29430f4cd3859b0be3e27575e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-30 14:23:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"755fed691d7ebef09cd19ab12b3af50516365467","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 19:47:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/06/781106/4/check/iscsi-driver-primera/9a01c30/9a01c30.html : FAILURE in 1h 00m 20s","accounts_in_message":[],"_revision_number":4},{"id":"7684a49f819e716089ef84d9ef6121d317d82136","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 19:48:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/06/781106/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/070d0c0/070d0c0.html : FAILURE in 1h 02m 08s","accounts_in_message":[],"_revision_number":4},{"id":"253d320285bae4f74bcd221289ee4bbf9c4a87ce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-30 20:16:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/781106/4/check/dsvm-tempest-cisco-zonemanager-job/7d5cf64/ : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":4},{"id":"d31c8841a390f723f3045b332bcde7258591da5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 22:25:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/274070470a15431da91cd9ac18dc45f1 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7bd8a3bdb774efb9b3dc4d06581b9cf : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3585511c62744f5a6e56dd4c4ddc738 : SUCCESS in 12m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5193b968d9c4b799ddcc85a37d6c59e : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebb705c7cdd54cd3b2d57b373998f242 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef9fa6702484e63807170c275080276 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/e0f20b88a61d4b8ca4075476c3930e7e : SUCCESS in 1h 16m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40d33991f9614599a7ab939cd8c34926 : SUCCESS in 1h 18m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9cf6f2d93414eb39cc4d522d3cb9d7a : SUCCESS in 14m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5349e9a395e641c28b070661f6677013 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5efcf18ad24b4ecf956d78bbe678b9d8 : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4718a4106e844a98bc39a2fb3e17c8e9 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e2cfc0160a244ae84aaaff5cc4b6550 : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3200046cb41b40ceb5033f26d72281a0 : FAILURE in 8m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/33d9d85c911244f9a5259ceb7a85b62a : SUCCESS in 1h 20m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ffb657f4eeb5480ab5870c9e2e657979 : FAILURE in 1h 41m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8432beb4384e447e8926eb5366e2f519 : SUCCESS in 1h 26m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5439caeb6e641ffaa2ca85ce1c10a21 : SUCCESS in 1h 18m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1afc1df2d5124b67b35ff00299d5159b : SUCCESS in 1h 26m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8d3f8b9c8eea46a7bc82e942f920be7e : SUCCESS in 1h 21m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e96286695bbd420486acd0a3070f53bb : SUCCESS in 2h 04m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e55b70fdd9624fd389dada2588ce800a : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":4},{"id":"4a78b58d5899019f57d638ecd80a9c058a217732","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-30 23:56:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/06/781106/4/upstream-check/cinder-cDOT-iSCSI/0932ed1/ : SUCCESS in 42m 58s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/06/781106/4/upstream-check/cinder-cDOT-NFS/d9ad889/ : SUCCESS in 42m 06s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/06/781106/4/upstream-check/cinder-cDOT-FCP/7a48613/ : SUCCESS in 1h 29m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"af74730d36a20aed90e20df5dd67d1ba8e87ca48","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-31 00:09:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/06/781106/4/upstream-solidfire-check/cinder-solidfire-iSCSI/fa44645/ : FAILURE in 1h 40m 56s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"58ead2b6834b5039e9b0383e76e8ccaa7f2a39a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-31 01:10:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/06/781106/4/check/iscsi-driver-nimble/a380819/a380819.html : FAILURE in 1h 17m 47s","accounts_in_message":[],"_revision_number":4},{"id":"eff5d89d37821c65987eec9bb594ed28d8ce9253","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-31 02:13:24.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781106-4/Lenovo-iSCSI : SUCCESS in 45:14.88\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781106-4/Lenovo-FC : SUCCESS in 42:41.42","accounts_in_message":[],"_revision_number":4},{"id":"04ba8564b6d4d609d5959437424c7c14ebb9b8a2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-31 02:31:52.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_SC_FC/df9eadc/DellEMC_SC_FC/547 : SUCCESS in 2h 04m 38s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_SC_iSCSI/af5edf8/DellEMC_SC_iSCSI/544 : SUCCESS in 2h 03m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"a3796536d6f19c8a66c1726d2ca43850bb388cbc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 04:01:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/06/781106/4/check/DellEMC_PowerFlex/855ca32/DellEMC_PowerFlex/697 : FAILURE in 2h 56m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c307ebb13572662439819240c415892db7197c1d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-31 06:16:00.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-4/ns5_nfs : FAILURE in 54m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-4/ns5_iscsi : FAILURE in 5h 38m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"df5235678889d9357a674f7d936aac708f265b68","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 09:38:02.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_XtremIO_FC/e66a495/DellEMC_XtremIO_FC/1193 : FAILURE in 2h 04m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_XtremIO_iSCSI/ea4075e/DellEMC_XtremIO_iSCSI/1211 : SUCCESS in 1h 45m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"1bebd41dfc11cf6718fa4c370bf6997c8b2cbb16","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-31 09:58:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_Unity_FC/69acbdf/DellEMC_Unity_FC/1038 : SUCCESS in 1h 43m 42s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_Unity_iSCSI/32e6b1c/DellEMC_Unity_iSCSI/979 : SUCCESS in 2h 09m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"4c251b8f7950bef344a7530b5ae80e93e01a3e13","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-31 11:11:28.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781106-4/PVME-iSCSI : SUCCESS in 56:00.92\n* PVME-FC http://os-logs.tristero.net/dell-ci/781106-4/PVME-FC : SUCCESS in 56:07.21","accounts_in_message":[],"_revision_number":4},{"id":"12396d1b8507c94b1c3a4593c4f37b432bff67f4","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-31 14:35:20.000000000","message":"Patch Set 4: Code-Review+2\n\nStill looks good to me.","accounts_in_message":[],"_revision_number":4},{"id":"2d93505adb26bd001ab6cd2b0985d72f147d6476","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-31 15:00:02.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781106-4/Seagate-iSCSI : SUCCESS in 47:49.15\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781106-4/Seagate-FC : SUCCESS in 39:52.18","accounts_in_message":[],"_revision_number":4},{"id":"96d9936e25feb2a4e6fa4877016ab9806a8955f9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-03-31 18:00:06.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/06/781106/4/fc : SUCCESS in 0h 51m 52s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/06/781106/4/iscsi : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":4},{"id":"01bb48cb26170edb6452d2cae1cef6b3ae76b03b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-31 18:15:25.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-4/ns5_nfs : SUCCESS in 6h 2m 42s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-781106-4/ns5_iscsi : SUCCESS in 5h 41m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c86d06ac18744af3bd440e9c5f46822521c39761","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 23:37:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_VNX_iSCSI/04215ea/DellEMC_VNX_iSCSI/679 : SUCCESS in 3h 24m 46s\n- DellEMC_VNX_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_VNX_FC/cd8541f/DellEMC_VNX_FC/728 : FAILURE in 1h 44m 47s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"4ea444a23e8f7090d89350419e494ab4f1ae3d1f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-31 23:46:43.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781106-4/HPMSA-iSCSI : FAILURE in 1:06:39\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781106-4/HPMSA-FC : SUCCESS in 33:26.57","accounts_in_message":[],"_revision_number":4},{"id":"190736465aa9a485efb1192d6175e7e16fd12aae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-01 13:21:52.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"33eb25c38425bee1ac4700cf30c1ffe53151698e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:23:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d05b9ba328ee36f128e75d16b72db2f6138eff0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:39:20.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c1a68aa5d8c3ffbb7b558f44cdcf9ce5484e4944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:00:51.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"df4e3183025e7c13e24095dec87c2215b51a1de1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-01 19:05:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/06/781106/4/20210402014100/NEC-ISCSI : SUCCESS in 2h 16m 50s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/06/781106/4/20210402014100/NEC-FC : SUCCESS in 2h 24m 55s","accounts_in_message":[],"_revision_number":4},{"id":"c2b56596f659a3cb5457d45022b1394fd71b9cca","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-01 19:06:14.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/781106/4/check/tempest-dsvm-full-focal-py3/22224aa/ : SUCCESS in 1h 43m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/781106/4/check/devstack-plugin-ceph-dsvm-focal-py3/9729c48/ : RETRY_LIMIT in 22m 12s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud finger://zuul-v3-executor.novalocal/29ee339443464b94a976206041814766 : DISK_FULL in 1h 34m 18s","accounts_in_message":[],"_revision_number":4},{"id":"d38654c19f35549b924a6268009edfcb673013db","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-02 03:37:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f4134365852fae16441d9ead717b282da7f90923","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 04:03:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37591776088e43acb72a1c4229a1db1b : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b9934e5b51d46179bfe33ca8ef43164 : SUCCESS in 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c082e08bbd382d2036a49d255ca77a56c8ed64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 05:36:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c203d89cf2264e46b2b1c1ed147891fa : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46ba4b672704ef896a60fc53a7eb7b2 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c976db15dd444cf8b4dae0471fc12eea : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f77a237141c54021873ddc0d44896c6e : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e61261c87b947eb963fafc0332aa5ee : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c597243f2646d6823e3d5740492226 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/659b7998d4084e198a9a5fbcf35dbda9 : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23bfaab5226a45a6b6c01d9fabd2fbad : SUCCESS in 1h 05m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1a97454383c2425a8e448c1e4114cab7 : SUCCESS in 11m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c42f956968f4e60a0d2f30460d351f8 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/098494a8acbc40379a53f033be53d359 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07eea15fc6d84012b4697b12a8fd935d : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb995e3edc774392b8a4f06f1ed8d412 : SUCCESS in 39m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f566a2811e754b558f2947c353b90635 : FAILURE in 10m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/53f6985b20734c299dd29fc05b5bbb7e : SUCCESS in 1h 13m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c41429c8be414a8bb86dc0849fadc07f : FAILURE in 1h 34m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8437bd5069a3489ca2ce7fb283577807 : SUCCESS in 1h 48m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/960912818a724f5b87857e03ba6a5a4b : SUCCESS in 1h 11m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e3e5ab7d7ddf4d40b3d23f71754b567e : FAILURE in 1h 29m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/07e7f0f69bfc475bba04841a95e48742 : SUCCESS in 1h 31m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2794e1a94d8d4fbdb66fba9be9e7d554 : SUCCESS in 1h 57m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff1f9e73a7034e2f9f7180f1142ce5c2 : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":4},{"id":"d85e924366c62e9838f5e8eadb2fe3e245de26f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 06:06:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c1c30a86fbde253e957a3ebe3612b4bde4461a0c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-02 07:24:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/06/781106/4/upstream-check/cinder-cDOT-iSCSI/e34198f/ : SUCCESS in 38m 12s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/06/781106/4/upstream-check/cinder-cDOT-NFS/7a7cd88/ : SUCCESS in 43m 41s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/06/781106/4/upstream-check/cinder-cDOT-FCP/3a8b434/ : SUCCESS in 1h 46m 21s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"43589669c7979dca63eba2aeaa0950eadb60fca7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-02 07:40:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/781106/4/check/dsvm-tempest-cisco-zonemanager-job/b89de9a/ : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9f942382f9f4cff4e59b5c5be854f2bc763b5549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 08:09:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76d9ea2ed88a4fd5af0d5ed004322dba : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca9c9e09aa74609ab32a8b98bf35f56 : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6cf6864c6714cb3bcd39b2e2d50ca3a : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62eaacf8ba5a47528d42a15a78cd8239 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d29af8e9834ad3941eea6c2cf78f95 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/f7ea5e9585f84944a0855b256939f8ab : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9247bb9d97764b01acaa8d33f4d1eb26 : SUCCESS in 1h 07m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d9d3f46e2b6d493c81b22a5876c019d0 : SUCCESS in 1h 14m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f27b5b8f3ce4862a58d9ce15e509ecb : SUCCESS in 1h 16m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96fe418df6684a52b9875ee070e73138 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":4},{"id":"34630eac23287992e7c0524490db5728a5f428cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 08:09:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"06dc7e642163999ed66ef97ca8da761e56d1394b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-02 08:49:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/06/781106/4/upstream-solidfire-check/cinder-solidfire-iSCSI/3905633/ : FAILURE in 55m 29s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"988c50ffdab3c8feada2bac2eb4ffa0eeb52ae7f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-02 10:12:26.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/781106/4/check/tempest-dsvm-full-focal-py3/aec2b26/ : SUCCESS in 1h 49m 43s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/781106/4/check/devstack-plugin-ceph-dsvm-focal-py3/b280c8a/ : SUCCESS in 1h 46m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3d60fbfe5e9ca1ce0ee8f2671883374e1c77970c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-02 13:04:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_SC_FC/fb4bda2/DellEMC_SC_FC/571 : SUCCESS in 2h 14m 33s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_SC_iSCSI/10eacb7/DellEMC_SC_iSCSI/568 : FAILURE in 2h 12m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"d78607c7f78ee767b8c7fa18a336a4a528dba58a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-02 20:32:33.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_XtremIO_FC/c2aa02e/DellEMC_XtremIO_FC/1222 : FAILURE in 2h 10m 39s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_XtremIO_iSCSI/ab7be75/DellEMC_XtremIO_iSCSI/1241 : SUCCESS in 1h 46m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"2189bb7f559be95903b1db1bdf7f861aa1e8e660","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-02 20:40:59.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_Unity_FC/6b7c5ec/DellEMC_Unity_FC/1070 : FAILURE in 1h 54m 56s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_Unity_iSCSI/3e8446a/DellEMC_Unity_iSCSI/1008 : FAILURE in 1h 55m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1dc8f1d24bba1a77deee54f641a7b5db5cd49929","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 00:37:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/06/781106/4/check/DellEMC_PowerFlex/e248451/DellEMC_PowerFlex/720 : FAILURE in 2h 51m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"870ee68c21ce1c87936b2845f91d27aacd0c0f8c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 01:35:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/06/781106/4/check/DellEMC_VNX_iSCSI/fed0d20/DellEMC_VNX_iSCSI/700 : FAILURE in 1h 18m 37s\n- DellEMC_VNX_FC http://publiclogs.emc.com/06/781106/4/check/DellEMC_VNX_FC/44104fa/DellEMC_VNX_FC/749 : SUCCESS in 1h 33m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"5dc3a6e08fb230a4ad12aaac292a04f940be9eb8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-04-03 14:33:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/06/781106/4/ : FAILURE in 51m 23s","accounts_in_message":[],"_revision_number":4},{"id":"0dfd8fba45d75d2c4d0100e265cead3b494b8546","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-04-03 20:00:05.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/06/781106/4/fc : SUCCESS in 0h 51m 52s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/06/781106/4/iscsi : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":4},{"id":"bb2a57c668c0498f7a0ddd07f09f21caa154c699","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-05 22:31:00.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781106-4/Lenovo-iSCSI : SUCCESS in 47:45.81\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781106-4/Lenovo-FC : SUCCESS in 43:14.68","accounts_in_message":[],"_revision_number":4},{"id":"12aa8522819ba5eb238ecb0b740f71b02864163f","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-06 00:01:31.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781106-4/Seagate-iSCSI : SUCCESS in 44:09.96\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781106-4/Seagate-FC : SUCCESS in 37:42.66","accounts_in_message":[],"_revision_number":4},{"id":"ba5a7bebf35b767b78da83723c66d84cced7dabc","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-06 11:33:18.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781106-4/HPMSA-iSCSI : FAILURE in 1:02:04\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781106-4/HPMSA-FC : SUCCESS in 34:10.94","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"935ddf7c7211c26141863d4579f40a04e5e3614e","revisions":{"30dde7be403d9e08daafcb66341be8aba4bc971d":{"kind":"REWORK","_number":1,"created":"2021-03-17 14:00:25.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/06/781106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/781106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/781106/1"}}},"commit":{"parents":[{"commit":"26bfba09845248e3223ffa4bae08b6a31db13f9c","subject":"Update cinder manage quota commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26bfba09845248e3223ffa4bae08b6a31db13f9c"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:00:31.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-17 09:02:00.000000000","tz":60},"subject":"Prevent quota and reservations to go into negative","message":"Prevent quota and reservations to go into negative\n\nOur current quota usage code doesn\u0027t prevent our usage or reservations\nto go into negative, though it warns at reservation time.\n\nThis patch makes sure that we don\u0027t go into negative, setting it to 0 in\ncase it would be negative.\n\nChange-Id: I487157d7092fa4cee1cd72cf01ce7031c94f553f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30dde7be403d9e08daafcb66341be8aba4bc971d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30dde7be403d9e08daafcb66341be8aba4bc971d"}]},"branch":"refs/heads/master"},"16f561c7dd8b6388c393791681d28bbe7762d8db":{"kind":"REWORK","_number":2,"created":"2021-03-26 11:26:48.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/06/781106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/781106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/781106/2"}}},"commit":{"parents":[{"commit":"a8d5a8ff7bb6539bd528d3d9619a290663f7ba4e","subject":"Update cinder manage quota commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8d5a8ff7bb6539bd528d3d9619a290663f7ba4e"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:00:31.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-26 11:26:15.000000000","tz":60},"subject":"Prevent quota and reservations to go into negative","message":"Prevent quota and reservations to go into negative\n\nOur current quota usage code doesn\u0027t prevent our usage or reservations\nto go into negative, though it warns at reservation time.\n\nThis patch makes sure that we don\u0027t go into negative, setting it to 0 in\ncase it would be negative.\n\nChange-Id: I487157d7092fa4cee1cd72cf01ce7031c94f553f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16f561c7dd8b6388c393791681d28bbe7762d8db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16f561c7dd8b6388c393791681d28bbe7762d8db"}]},"branch":"refs/heads/master"},"fb6a26f312be5895c1ea81354df1a15f7a19d552":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-30 13:48:21.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/06/781106/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/781106/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/781106/3"}}},"commit":{"parents":[{"commit":"8f78194143ec7d955b09638b1942866cfdaf3b22","subject":"Update cinder manage quota commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f78194143ec7d955b09638b1942866cfdaf3b22"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:00:31.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-30 13:46:39.000000000","tz":120},"subject":"Prevent quota and reservations to go into negative","message":"Prevent quota and reservations to go into negative\n\nOur current quota usage code doesn\u0027t prevent our usage or reservations\nto go into negative, though it warns at reservation time.\n\nThis patch makes sure that we don\u0027t go into negative, setting it to 0 in\ncase it would be negative.\n\nChange-Id: I487157d7092fa4cee1cd72cf01ce7031c94f553f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb6a26f312be5895c1ea81354df1a15f7a19d552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb6a26f312be5895c1ea81354df1a15f7a19d552"}]},"branch":"refs/heads/master"},"935ddf7c7211c26141863d4579f40a04e5e3614e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-30 14:23:56.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/06/781106/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/781106/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/781106/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/781106/4"}}},"commit":{"parents":[{"commit":"cdd60605f45f9ed624150d1310d8bce2cbef48c3","subject":"Update cinder manage quota commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdd60605f45f9ed624150d1310d8bce2cbef48c3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 16:00:31.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-30 14:23:10.000000000","tz":120},"subject":"Prevent quota and reservations to go into negative","message":"Prevent quota and reservations to go into negative\n\nOur current quota usage code doesn\u0027t prevent our usage or reservations\nto go into negative, though it warns at reservation time.\n\nThis patch makes sure that we don\u0027t go into negative, setting it to 0 in\ncase it would be negative.\n\nChange-Id: I487157d7092fa4cee1cd72cf01ce7031c94f553f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/935ddf7c7211c26141863d4579f40a04e5e3614e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/935ddf7c7211c26141863d4579f40a04e5e3614e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
