)]}'
{"id":"openstack%2Fcinder~781515","triplet_id":"openstack%2Fcinder~master~I90e9f85d192e1f2fee4ec8615a5bc95851a90f8e","project":"openstack/cinder","branch":"master","topic":"quota-retype","hashtags":[],"change_id":"I90e9f85d192e1f2fee4ec8615a5bc95851a90f8e","subject":"Change snapshots type with volume retype","status":"MERGED","created":"2021-03-18 17:02:42.000000000","updated":"2021-04-04 11:48:08.000000000","submitted":"2021-04-01 09:53:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781515-quota-retype","meta_rev_id":"f7aae2aa1302455171cfbc69df93d00e9db5a3c0","_number":781515,"virtual_id_number":781515,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 09:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-31 18:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-31 13:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-31 18:07:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-31 18:07:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 17:11:02.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-18 17:29:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 17:47:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 18:03:39.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-03-18 18:32:29.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 19:27:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-18 20:06:13.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 20:24:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 20:37:39.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 22:25:53.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-20 22:59:14.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 18:57:23.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-22 03:10:06.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 15:41:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 16:42:57.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-26 20:07:21.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 22:57:35.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-27 17:19:08.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-28 05:15:48.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-29 10:09:38.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-29 13:53:54.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 01:57:52.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-31 13:14:14.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-31 18:07:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-04-01 09:09:19.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-04-01 22:05:24.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 03:34:44.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-04 11:48:08.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ecb09b8f3be34b6b20052a12767614704d9b5055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-18 17:02:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7b58674823a0ea21b190e00501e69b1859da95f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-18 17:11:02.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/15/781515/1/check/dsvm-tempest-TOYOU-ACS5000-driver/0500624 : FAILURE in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"aebce881efde084adf1c2be7a68d83b63ec34dd7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-18 17:29:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/15/781515/1/check/iscsi-driver-primera/21be2af/21be2af.html : FAILURE in 25m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ca4e736f6aef0476962bf3532d31830464220373","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-18 17:43:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/15/781515/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c74b527/c74b527.html : FAILURE in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1395f096a5823519284ee1d71dc70594766a640c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-18 17:47:14.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781515/781515/1/check-not-vote/ext-cinder-zuul/00cbad8 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8ddc7ca8b977e3578508ba54390ffe0c12f815b4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-18 18:32:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/15/781515/1/ : SUCCESS in 1h 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5b115da414eab2ca4a2e4ca3b6266d35578f8024","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-18 19:27:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/15/781515/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e60dda2/ : SUCCESS in 2h 22m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/15/781515/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d329a93/ : SUCCESS in 2h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2584d5299fa47afc63729847f6966134bb6efce0","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-18 20:06:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/781515/1/check/dsvm-tempest-zadara-driver/05c2cd6 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"986bf048ed25a26dcd78db4507098a79631ee660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:24:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8ff9ff473984af8b08d43924dee73dc : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07c80e2cd934da5880610ad19a30269 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c5de32cfcaf44889cfd1519e72c1b29 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5831993b60874b009a025087651101a9 : SUCCESS in 11m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aca639d60cc04d74b6b1c6578047eaea : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782a3b1ff5de4ee1be91d574a7a6381c : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/01132d5322da4b4693cbcf5896dea2cf : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a605801415af44279153731acf38932c : SUCCESS in 1h 17m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc2765ce98d547c6beb95714f76d7527 : SUCCESS in 7m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/af0f5d6ce5e74ca9983e7f0018f551b6 : SUCCESS in 15m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd1109e7d09546d4884a63966a5ee9d7 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/350c48f3680940e5b7fb52040fcd3ea5 : SUCCESS in 7m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4233a3687824a5f955964133d6dbd13 : SUCCESS in 7m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e89db197f2e9472fb8158fccfeb0f704 : SUCCESS in 43m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d41b2e49791402a847f9e7f4b0f8b3a : FAILURE in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64c849fce6ef4aac8a0cfc0d973de2a9 : SUCCESS in 1h 21m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65503ff68f234d89b607d2cdde470211 : FAILURE in 1h 27m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b267e7c11134204ac3e53266d864034 : SUCCESS in 1h 39m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/48f020dcaacc47ffb55487065d9854d1 : SUCCESS in 1h 23m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a569749da88f4f51a466875ac278d8ab : FAILURE in 1h 38m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ee433f085ef74f478f011a9e09806968 : SUCCESS in 1h 57m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbc5d1cd30f44eff90fcf5327fa4307d : FAILURE in 2h 59m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71897a06b3124c14bf1ba3db209f2ae7 : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c807baa73a23e142a713e76f960ccb1a8c04f213","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-18 20:37:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/1/check/tempest-dsvm-full-focal-py3/a463c02/ : SUCCESS in 2h 27m 42s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/1/check/devstack-plugin-ceph-dsvm-focal-py3/f892d5a/ : SUCCESS in 2h 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5250b6cf6dca4b0525d1f59e2f1768dc29087487","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-18 22:25:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/781515/1/check/dsvm-tempest-cisco-zonemanager-job/d0bfde4/ : FAILURE in 1h 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"eb9a7bf8435fa0a32864a530e3c8db8fe45123b5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-20 13:30:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/15/781515/1/check/iscsi-driver-nimble/400fded/400fded.html : FAILURE in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"87f8419cd4fb87052c880c4af6279054427ca3c7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-20 22:59:14.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-781515-1/ns5_nfs : FAILURE in 5h 36m 0s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-781515-1/ns5_iscsi : FAILURE in 5h 33m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4f50a8893821e1d3a13cc71544b5c2b4af01b56a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-21 18:57:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/781515/1/check/DellEMC_Unity_FC/a0ffc7a/DellEMC_Unity_FC/921 : FAILURE in 2h 27m 43s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/781515/1/check/DellEMC_Unity_iSCSI/9488bce/DellEMC_Unity_iSCSI/873 : FAILURE in 2h 35m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9091d40a6c0a4efca792c1c4b53c317da67dfe62","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 03:10:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/781515/1/check/DellEMC_XtremIO_FC/4a527de/DellEMC_XtremIO_FC/1086 : FAILURE in 2h 22m 41s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/781515/1/check/DellEMC_XtremIO_iSCSI/a2278ed/DellEMC_XtremIO_iSCSI/1090 : FAILURE in 2h 03m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"7987f70af696bc86c03a36837627ceb0a0468c73","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 14:57:22.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-781515-1/ns5_nfs : FAILURE in 5h 7m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-781515-1/ns5_iscsi : FAILURE in 5h 15m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"99fbedbd433d9ef9e2ffc7689c3387de4354ed14","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-22 15:01:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51e3ec53d4d96d23797df80c5353942ad4578f21","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 15:41:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/781515/1/20210322221801/NEC-FC : SUCCESS in 2h 20m 44s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/781515/1/20210322221801/NEC-ISCSI : SUCCESS in 2h 23m 15s","accounts_in_message":[],"_revision_number":1},{"id":"97cdd74c12c123fcfcead360821c8ea7f807107f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-22 15:43:29.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781515/781515/1/check-not-vote/ext-cinder-zuul/923f140 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2fb5a01c1872e05d4e5e34e29c166f8699cc1803","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-22 17:36:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/781515/1/check/dsvm-tempest-cisco-zonemanager-job/fcf6eb2/ : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":1},{"id":"af166c5e09fb0a033890f830e0c054c29c91392f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-22 18:10:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/15/781515/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/17c96e9/ : SUCCESS in 2h 23m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/15/781515/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d8b901e/ : SUCCESS in 2h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5a2854a484edcfc9cafc81ee7ed2e8ab8fae7805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:38:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fe263b4b61a4440b8c5e5d96efe9028 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50ebea520d24188892bea0aa168405d : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69eaed6a99b7444c9f075f7638aa2c62 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e24ec9846a44c92be812607fed01b17 : SUCCESS in 17m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e32f9d5ed7542a792a272e45fc01726 : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b53b6655634dada7b0712d1ce6faf1 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/d1b325ad9351429d8085923f530f2cbc : SUCCESS in 1h 09m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c1687fa499845e69b2e1688a64a4dce : SUCCESS in 1h 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaffb47ee4ae41888944b1fd24880800 : SUCCESS in 6m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b5befa6ce48e439b99dd302eef222514 : SUCCESS in 12m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6dbffa519b154bc7a064379eba99ebe2 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68aff94ec1044341b7c25a755690d052 : SUCCESS in 6m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94d5bec0bc384b82a9f6d017e8896a5d : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf125a133d094663b8a45bfb22672e5a : SUCCESS in 44m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b978a5242b14a44867565384d85292f : FAILURE in 12m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a305643c82cc4c839f5201d1b7ba4424 : POST_FAILURE in 55m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c961985c21dc482b85bb13d83de68f7c : FAILURE in 1h 37m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb9551d1544042c594e6e72e7efca3c8 : SUCCESS in 1h 30m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/59367a54bf3442e4921f644d58b818c8 : SUCCESS in 1h 19m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9729c0f5e0c4e1281eb0b33786b9c6b : FAILURE in 1h 35m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/175141755f7a420d9e5f526f3c91174a : SUCCESS in 1h 11m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51f05897a3f84c60916680f4dcd94281 : FAILURE in 1h 48m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29db481b840c4472b5c9bc33926e1373 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ff51203dee57bac144a1903329c1e4794820899a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-22 21:09:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/781515/1/check/dsvm-tempest-zadara-driver/59c476e : SUCCESS in 1h 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"739cbafed26a5415fc90dc28b19ed9e71097bc92","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-23 04:52:41.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/1/check/tempest-dsvm-full-focal-py3/378a901/ : SUCCESS in 1h 58m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/1/check/devstack-plugin-ceph-dsvm-focal-py3/c4279c3/ : SUCCESS in 2h 22m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a9dcfa5faf767456069a1697516e8bbbf44d1571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-26 12:57:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4f5d3c43bb840d9a6d8b616c2a31c853b6495d5c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-26 13:01:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"69f2c0699f28d2d5ea5a646f0a0232062e4c8343","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 13:24:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e986c0ec5174a868c642a6dcb7a9cd8 : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f389401e98e849819f18fc56985d318b : SUCCESS in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a53cfe13ebab35bb88740062536449eb244a656f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 15:08:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/145838e168f945f9bbb35e77be191237 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f501f4d4db9c4d13864b68b9560830f0 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2e7ddb4a64a4b13afa0a759d28744d9 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8acafa6b3d9471285ac94e074c50cc0 : SUCCESS in 13m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45aeacc4b3d944a6b201d377041e3b9c : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8f1b005bab486986a317514b8ace66 : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/4cddc5c33f8445d683a405a0209d3895 : SUCCESS in 1h 07m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c40bbca90e746e68401564e968f9b9a : SUCCESS in 1h 06m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/467c9f1011d54c17bd4e78a20ef36a73 : SUCCESS in 7m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e3b734f6c50f4f71b478f3da8451e780 : SUCCESS in 14m 22s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88fa40eef82a4f33b70f5b77830ef790 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a53f568b9cca4652841d876d648b40b9 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2791ed72d874222a13326c246529c49 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dedb894611aa4f99adcf2df5f5d5c8f3 : SUCCESS in 42m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14d8aa5e9142402695986e39f082fd68 : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/529d01e2e93d45fbad08ae356a04ec4a : SUCCESS in 1h 22m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cbb0b088fa2240d7ab22b21997151fb0 : FAILURE in 1h 25m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1fa1605d86dd4ab385e79e8f751e5eb4 : SUCCESS in 1h 17m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ac8ee2d90da4bf786e4cecf09af683b : SUCCESS in 1h 29m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a24351e3c30a4eb3bcb99ee43dbd256e : SUCCESS in 1h 27m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7976fbe5936a425795d2c173e422a06f : SUCCESS in 1h 27m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58a1c61a94d3420e8c3dccc6673d7116 : SUCCESS in 2h 04m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f4cc5d58524843a4835fdf48dfeee9 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":2},{"id":"01c98f7beaca60b6c3deec6b8bae9eeb15e8639a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-26 16:42:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/15/781515/2/upstream-solidfire-check/cinder-solidfire-iSCSI/b9dee7a/ : FAILURE in 54m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"44c1920112372c726689657dc7fd0812e7ef2f07","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-26 20:07:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/15/781515/2/upstream-check/cinder-cDOT-iSCSI/ee48af6/ : SUCCESS in 41m 40s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/15/781515/2/upstream-check/cinder-cDOT-NFS/953bfb6/ : SUCCESS in 3h 11m 52s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/15/781515/2/upstream-check/cinder-cDOT-FCP/a1c194c/ : SUCCESS in 1h 09m 14s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"880fa51ab0828fd9e75c0771137e1bcdde8ffac2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-26 22:57:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_PowerStore_iSCSI/f2180bb/DellEMC_PowerStore_iSCSI/728 : FAILURE in 38m 47s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_PowerStore_FC/fe23b70/DellEMC_PowerStore_FC/722 : FAILURE in 36m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"0e421ae7ac67a262e58d1eedb4d70dac8cd02730","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-27 03:18:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/781515/2/check/dsvm-tempest-cisco-zonemanager-job/fc70944/ : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":2},{"id":"68cf1327ecb2b2988aa3ca3bd759557c849ce9d3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-27 09:40:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/781515/2/check/dsvm-tempest-zadara-driver/7589848 : SUCCESS in 1h 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec85d138139247f438ab88e12ee154950101941e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-27 12:21:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-781515-2/ns5_nfs : SUCCESS in 5h 27m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-781515-2/ns5_iscsi : SUCCESS in 5h 22m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"2978962ffda126699972ca51cb31489bd65563f4","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-27 17:19:08.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781515-2/Lenovo-iSCSI : SUCCESS in 45:44.93\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781515-2/Lenovo-FC : SUCCESS in 43:35.27","accounts_in_message":[],"_revision_number":2},{"id":"ba90aac93e4a88399d5d0e94da95f52b927bde0d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-28 02:40:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/15/781515/2/check/iscsi-driver-nimble/0215354/0215354.html : SUCCESS in 1h 50m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9e97cb257fede8a1734c64b7146d690b623266e9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-28 05:05:24.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cdbcaf04ed0a47c897235e544385806c : POST_FAILURE in 1h 07m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/1a11dd8a520b481e8895d5e70fbef673 : POST_FAILURE in 1h 16m 11s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/2/check/devstack-plugin-ceph-dsvm-focal-py3-newcloud/bb82a08/ : FAILURE in 1h 21m 18s","accounts_in_message":[],"_revision_number":2},{"id":"cbbb11120228e846f50a9a972699a445f4aa2003","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-28 05:15:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_SC_FC/453cc28/DellEMC_SC_FC/527 : SUCCESS in 2h 00m 08s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_SC_iSCSI/620f9b7/DellEMC_SC_iSCSI/524 : SUCCESS in 2h 20m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"2cd960ce1138c1254e87bb4122df7e719a25e0ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-28 12:17:16.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_Unity_FC/e6b6950/DellEMC_Unity_FC/1009 : SUCCESS in 1h 30m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_Unity_iSCSI/3d725ea/DellEMC_Unity_iSCSI/950 : SUCCESS in 1h 51m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d12473f317d2d2c6dd556838e32e2d74868bfb8f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-29 00:24:27.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_XtremIO_FC/e8ba3fe/DellEMC_XtremIO_FC/1164 : SUCCESS in 2h 01m 52s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_XtremIO_iSCSI/d7c03c5/DellEMC_XtremIO_iSCSI/1182 : SUCCESS in 1h 43m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"0e92b08217dfd1f35648526027e717f585165be5","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-29 10:09:38.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781515-2/PVME-iSCSI : SUCCESS in 56:37.80\n* PVME-FC http://os-logs.tristero.net/dell-ci/781515-2/PVME-FC : SUCCESS in 56:18.30","accounts_in_message":[],"_revision_number":2},{"id":"bd83897c8621a23424c7cbb6e682a14c25c01e17","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-29 13:53:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_VNX_iSCSI/7821cad/DellEMC_VNX_iSCSI/662 : SUCCESS in 3h 21m 52s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_VNX_FC/47c854d/DellEMC_VNX_FC/708 : SUCCESS in 1h 27m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2809b38370821f8439de7743374fb0dc88417759","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-30 01:57:52.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781515-2/Seagate-iSCSI : SUCCESS in 45:59.71\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781515-2/Seagate-FC : SUCCESS in 38:03.24","accounts_in_message":[],"_revision_number":2},{"id":"fc4786ba81f899983f2e906c8f0262570ddb6dcc","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 13:34:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/781515/2/20210330194028/NEC-ISCSI : SUCCESS in 2h 46m 24s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/781515/2/20210330194028/NEC-FC : SUCCESS in 2h 53m 44s","accounts_in_message":[],"_revision_number":2},{"id":"f216c3df9a4b788ecb9776d526a6e00014abc9e9","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-31 13:14:14.000000000","message":"Patch Set 2: Code-Review+2\n\nThis change makes sense to me!","accounts_in_message":[],"_revision_number":2},{"id":"e5748cdec82dd92d3786bc7dd8bec203774b3217","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-31 18:07:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1 Review-Priority+1\n\nCode looks good and has good unit test coverage.","accounts_in_message":[],"_revision_number":2},{"id":"0a3bcc77769f6b23b4ead9c478c991b2086e6380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:44:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b971e5ac13d02c9c84a6d9de8d7116b036090583","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-01 05:50:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a6d295666dfa277bb5d05c4adf74a4bbce4ca91d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 06:17:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e76c63bee22545e58b4beefdc139935e : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c1b1ab8a65743f29ee84512d0f59b19 : SUCCESS in 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d2b7bfc4dce9ba1f7afe8d56958c7d331da8547","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-01 07:18:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/781515/2/check/dsvm-tempest-cisco-zonemanager-job/1ccc2d9/ : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a6c96ab0a0377c2663a835de63a474e810ea1e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 08:13:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abf984685c5e4b50a2b6ee4ae148c749 : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b926b8a05ce4deabc345290849c08e4 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9451541493604efe9b8907760ad3c8f9 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9af285df7a694207a8333d37280b6808 : SUCCESS in 9m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f89c9043d80440958d676f64418c467a : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63913a064d5544c89aa1ab0e60b1c2e3 : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/6fe205b1b30c471699c242eb37be36c3 : SUCCESS in 1h 20m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f877950238734f36b57aee1c9eea8094 : SUCCESS in 1h 28m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb0de3e8edb4458e87b6e29f119a8492 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5894243f3f174f4380f16c6e54825c14 : SUCCESS in 11m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/589146dc4760433e9530ee161f424707 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae775f25c6414f1aac1bf2141f3a93b7 : SUCCESS in 8m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5b10eafd03a4675af19b9c4a94b523d : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/63c1f503c7d545538be8a809c1ff3712 : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7941bf4cf764b119e3a221e031a9ff3 : FAILURE in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4184b7f400c416b892ef58048f6b923 : SUCCESS in 1h 26m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b37bfe8d54748da968d6c29bf1e69f5 : FAILURE in 1h 37m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/29e0a51f103f491c877ac634c4d88027 : SUCCESS in 1h 50m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a70b3bbb424f4dbb964eff2be389e85f : SUCCESS in 1h 16m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/482fa51d31614b32bf7ebeb560ce3024 : SUCCESS in 1h 33m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0df54a4e386840f1aa55f84cfdf1502b : SUCCESS in 1h 13m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/286db92e6da9401997d56441137953f4 : SUCCESS in 2h 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34bd164fcfe045d5ae54380a33718a31 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7ede5405faf996e84ff80db7d8cefa49a91019b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 08:14:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8715f032b61fd48575439e13dd5fd2caed8e3a0c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-04-01 09:09:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-15-781515-2/1617264834 : SUCCESS in 28m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"e01d798a231a67982062247b7d85563d19056286","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-01 09:27:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/15/781515/2/upstream-check/cinder-cDOT-iSCSI/1cac999/ : FAILURE in 59m 08s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/15/781515/2/upstream-check/cinder-cDOT-NFS/1ba655c/ : FAILURE in 36m 39s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/15/781515/2/upstream-check/cinder-cDOT-FCP/4d7b8b5/ : FAILURE in 1h 13m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"aa60c290af085d6df878b75d55fc647f1c0b35d2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-01 09:40:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/15/781515/2/upstream-solidfire-check/cinder-solidfire-iSCSI/5cc4585/ : FAILURE in 1h 25m 48s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"d44a678fa91cb79eab2752c5c0397caa74a6e4f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:53:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/717684b3983a4c5a940dbd2abf7ca770 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9ca306a49d4b9a871d557ab56b4577 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89ddc30b030140bfbd9aa2202f361986 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef4b7fbd93d54238a543fd32232b9041 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763af0626e1a43a4bbc5f48b6b12e4b0 : SUCCESS in 17m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/09b986f03242476b91367d5b18086f64 : SUCCESS in 1h 19m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7462676973c84601885f0fd125c5efa5 : SUCCESS in 1h 25m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bc061a011304df085822cf77301d2b7 : SUCCESS in 8m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a4ef9c3f94a49b9a26d0f48ac337ad4 : SUCCESS in 1h 13m 45s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a409edf9122f4d13a9ae25b2f7000bad : SUCCESS in 1h 14m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8be434bb361436f9912c37944024b3b : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6f2c756c13b6f9d226e5d13a7f661963742b60d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:53:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"306ad11c2f2733056e3fbadadc9816e5cfaba766","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:00:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d37ffd4b8948f1a5206ba2d3bdb797 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f02770715114646b036627115a8fcaa : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2},{"id":"3c34bb47922552bc97369d1cd3060f6adaa6a143","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-01 12:24:19.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781515-2/Lenovo-iSCSI : SUCCESS in 47:50.34\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781515-2/Lenovo-FC : SUCCESS in 43:41.98","accounts_in_message":[],"_revision_number":2},{"id":"047034a6d73ebe0c66c25e209d0e47a84bc41296","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-01 12:45:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_SC_FC/b19a5e4/DellEMC_SC_FC/561 : FAILURE in 2h 19m 35s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_SC_iSCSI/b5dc2e3/DellEMC_SC_iSCSI/558 : FAILURE in 2h 11m 10s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"b0bc41263e0aee18ca828e991b25bd994cfae2f3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-01 13:14:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/781515/2/check/dsvm-tempest-zadara-driver/1847837 : SUCCESS in 1h 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98be1d171185ea1fa8f6cc2acb39290434688dd5","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-01 18:14:18.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781515-2/Seagate-iSCSI : SUCCESS in 45:27.94\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781515-2/Seagate-FC : SUCCESS in 39:22.02","accounts_in_message":[],"_revision_number":2},{"id":"f65ea56e5f6ac5ee731fab44ca023bb06e7a0ce1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-01 22:05:24.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/15/781515/2/check/DellEMC_PowerFlex/251e50d/DellEMC_PowerFlex/711 : FAILURE in 3h 15m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"050093df3a323620950bccda14f3c2c7c3dac69c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-01 23:05:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_XtremIO_FC/10f0a8d/DellEMC_XtremIO_FC/1211 : FAILURE in 2h 04m 28s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_XtremIO_iSCSI/e6ca72f/DellEMC_XtremIO_iSCSI/1229 : SUCCESS in 1h 46m 55s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"00284a1cfa084c85a6efe1e5cb555250a81b9059","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-01 23:45:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_Unity_FC/31926e7/DellEMC_Unity_FC/1058 : SUCCESS in 1h 43m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_Unity_iSCSI/52bb815/DellEMC_Unity_iSCSI/997 : FAILURE in 2h 07m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f5e8484f1684e8c08181299a966ebbb7649a7f8b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-02 03:34:44.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781515-2/HPMSA-iSCSI : FAILURE in 58:34.60\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781515-2/HPMSA-FC : SUCCESS in 34:47.72","accounts_in_message":[],"_revision_number":2},{"id":"17ec5a56a8667d2f4d35570cdc4be729cf5beaaf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-02 08:27:41.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/2/check/tempest-dsvm-full-focal-py3/f75969e/ : SUCCESS in 1h 38m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/2/check/devstack-plugin-ceph-dsvm-focal-py3/cd3f9b2/ : SUCCESS in 1h 52m 20s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/781515/2/check/devstack-plugin-ceph-dsvm-focal-py3-newcloud/a7ba0ad/ : TIMED_OUT in 3h 11m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b29d42b9ea2ee02c2e64740fbc5f127ab9703fb2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-02 15:26:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_VNX_iSCSI/8894a38/DellEMC_VNX_iSCSI/692 : FAILURE in 1h 14m 33s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_VNX_FC/09a6eea/DellEMC_VNX_FC/741 : SUCCESS in 1h 44m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f7aae2aa1302455171cfbc69df93d00e9db5a3c0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-04 11:48:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/781515/2/check/DellEMC_PowerMAX_FC/6d30460/DellEMC_PowerMAX_FC/673 : SUCCESS in 6h 55m 32s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/781515/2/check/DellEMC_PowerMAX_iSCSI/83890ac/DellEMC_PowerMAX_iSCSI/628 : SUCCESS in 5h 04m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b77fb72aa9a58668b2c39f8c9eff06f920d76a74","revisions":{"69bdf1998058237b2e1a7a5422a9caf6bb0cd056":{"kind":"REWORK","_number":1,"created":"2021-03-18 17:02:42.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/15/781515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/781515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/781515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/781515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/781515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/781515/1"}}},"commit":{"parents":[{"commit":"d1665ea99ebe30db8116f1f437c3eba8c0777b90","subject":"Merge \"[DS8K]: Support revert to snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1665ea99ebe30db8116f1f437c3eba8c0777b90"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-18 16:19:34.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-18 16:19:34.000000000","tz":60},"subject":"Change snapshots type with volume retype","message":"Change snapshots type with volume retype\n\nBoth volumes and snapshots have a volume_type_id field in the DB, but\nwhen we migrate a volume we leave the snapshots with the old type.  This\ncan only happen for retypes that don\u0027t do migration, since we cannot do\na retype with migration if the volume has snapshots.\n\nLeaving the snapshots with the old time makes a mess of the quota usage\nwhen we do the retype as well as when we delete the snapshots.\n\nThis patch fixes the quota issue by making sure the snapshots are\nretyped as well.  This means that we will check quotas for the snapshots\nwhen retyping a volume that have them and we will properly reserve and\nset the quota on retype.\n\nCloses-Bug: #1877164\nChange-Id: I90e9f85d192e1f2fee4ec8615a5bc95851a90f8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69bdf1998058237b2e1a7a5422a9caf6bb0cd056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69bdf1998058237b2e1a7a5422a9caf6bb0cd056"}]},"branch":"refs/heads/master"},"b77fb72aa9a58668b2c39f8c9eff06f920d76a74":{"kind":"REWORK","_number":2,"created":"2021-03-26 12:57:46.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/15/781515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/781515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/781515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/781515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/781515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/781515/2"}}},"commit":{"parents":[{"commit":"d1665ea99ebe30db8116f1f437c3eba8c0777b90","subject":"Merge \"[DS8K]: Support revert to snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1665ea99ebe30db8116f1f437c3eba8c0777b90"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-18 16:19:34.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-26 12:57:35.000000000","tz":60},"subject":"Change snapshots type with volume retype","message":"Change snapshots type with volume retype\n\nBoth volumes and snapshots have a volume_type_id field in the DB, but\nwhen we migrate a volume we leave the snapshots with the old type.  This\ncan only happen for retypes that don\u0027t do migration, since we cannot do\na retype with migration if the volume has snapshots.\n\nLeaving the snapshots with the old time makes a mess of the quota usage\nwhen we do the retype as well as when we delete the snapshots.\n\nThis patch fixes the quota issue by making sure the snapshots are\nretyped as well.  This means that we will check quotas for the snapshots\nwhen retyping a volume that have them and we will properly reserve and\nset the quota on retype.\n\nCloses-Bug: #1877164\nChange-Id: I90e9f85d192e1f2fee4ec8615a5bc95851a90f8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b77fb72aa9a58668b2c39f8c9eff06f920d76a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b77fb72aa9a58668b2c39f8c9eff06f920d76a74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
