)]}'
{"id":"openstack%2Fcinder~781758","triplet_id":"openstack%2Fcinder~master~Ie3c6367c01aaec6e5f97ccb17343022c9308c7e6","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2026-01-02 16:30:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie3c6367c01aaec6e5f97ccb17343022c9308c7e6","subject":"DNM Test NFS CI","status":"ABANDONED","created":"2021-03-19 14:58:21.000000000","updated":"2026-01-02 16:30:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9028fd25ef2a98ba3939faaebce585c9f53b747","_number":781758,"virtual_id_number":781758,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-19 22:49:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-03-22 17:35:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-03-22 17:35:28.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 15:07:58.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-19 16:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 16:28:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-19 16:39:49.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 17:12:00.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-19 19:12:36.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-19 19:30:27.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-19 19:34:05.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-19 20:14:42.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 01:24:49.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-20 03:37:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 11:14:07.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 05:42:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 07:16:19.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-21 16:52:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 18:43:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-23 03:53:55.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 13:22:17.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-27 16:55:39.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"}],"messages":[{"id":"121ead1c2f4106bff44bc96bcb5d8be92aad8fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-19 14:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"033342b85f06f73315ad4aba2f05bfbc667cd333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-19 15:00:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d02538cff360ba9168544ebb1b6c9aefaac50b1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-19 15:07:58.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/58/781758/2/check/dsvm-tempest-TOYOU-ACS5000-driver/8462d49 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f3106cfb4408b75abfb7fe86e3dc22efb1472752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:10:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eeb9454b5ee442059868bbf5ab81badf : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63fc29f57673414184ef627cbce55dc2 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f263d77b86654448a08ee893ed173136 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45d8bf108da94727a3ced56c80103868 : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581fc517f2bf4e20b136aa70aab56198 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2d5e70d206248c59e612b4a28eb3259 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b97f0574f3439097919c90b9c59767 : SUCCESS in 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8de9d0cf5a84699ab3391228b9af581 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3ccbe73b52040dcb229f4f6bc0cfbb5 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/473917b880374b2888529a9a19d94097 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/528799bd883c4a769e62bba9d112fd82 : SUCCESS in 42m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57c10a9a62224813bc15859c7d6cfddb : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45a5f6fb459b425ea2e41ee3cc0608dce32f58cc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-19 16:28:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/781758/2/check/dsvm-tempest-cisco-zonemanager-job/0bb8f99/ : FAILURE in 1h 22m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ea4278c81a09a47a2c19d326bf7415becf371f03","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-19 16:39:49.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781758/781758/2/check-not-vote/ext-cinder-zuul/fe0f169 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e6631a460469f55bd50cda7576f11e156fa133ea","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-19 17:12:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/781758/2/check/DellEMC_PowerStore_iSCSI/ce71a76/DellEMC_PowerStore_iSCSI/654 : SUCCESS in 1h 00m 51s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/781758/2/check/DellEMC_PowerStore_FC/2af727c/DellEMC_PowerStore_FC/648 : SUCCESS in 55m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5bf6b0fb7bc18112cead99a222be31cce83ffb18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-19 18:05:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07ac15cfd6804e9035dee73e8a2d4553681a3ae1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-19 18:10:47.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/58/781758/3/check/dsvm-tempest-TOYOU-ACS5000-driver/a7e2043 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3b1ad72c44bb21587d042474b31230b79830cdce","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-19 19:12:36.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781758-2/Seagate-iSCSI : SUCCESS in 45:17.54\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781758-2/Seagate-FC : SUCCESS in 39:23.20","accounts_in_message":[],"_revision_number":2},{"id":"641d7dff5e28ba9f51695bf958556ab1b5c3269b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-19 19:30:27.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781758-2/Lenovo-iSCSI : SUCCESS in 50:15.35\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781758-2/Lenovo-FC : SUCCESS in 46:09.99","accounts_in_message":[],"_revision_number":2},{"id":"8ee9b6b7aeebc330ab66f1684e7489c1f0e6e6f4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-19 19:34:05.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781758-2/HPMSA-iSCSI : FAILURE in 1:03:56\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781758-2/HPMSA-FC : SUCCESS in 34:19.71","accounts_in_message":[],"_revision_number":2},{"id":"472f89757d43ba0702cea2adb1c0d962a3edbfcd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-19 20:14:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/781758/3/ : SUCCESS in 1h 36m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c34de69ea7facb9abfbafaf21fbd8af3ffe20fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 20:33:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b2c291a1e9b54ab89bcd7f84700a84df : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b7f6c582c35443c965739b8c6d4939c : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5623829bb9e64e91ada9f148402090bd : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f785b7e2fa2b4a15bc51e28a9879ff38 : SUCCESS in 12m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04df05c9eacf49ed9372ca52e5ad220b : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/595daecd001a4accb05cc54324e27f66 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5471d7ea754c81b2cc21073bd7333e : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3ba05eb67ee543638d8fc07b4f5a62c8 : SUCCESS in 43m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e88c28b41744456dbd5c4546f7404f73 : SUCCESS in 1h 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/512ecb5757264299b54e9a726e959ad3 : SUCCESS in 10m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f41b54c518f4241aefcd261db079a51 : SUCCESS in 13m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f29318b8a03e4073a16eeca7eb3ab426 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dac63c7af95d448abc0238beb6759f2b : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d0285f72c8f49ca971ccd762536899d : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1d7e27c6e4442e69a747363dbbe9d22 : SUCCESS in 40m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9985afe0fd204b7d930091f5e0479797 : SUCCESS in 17m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e13bf04aa3db473288b4904165da85e6 : TIMED_OUT in 2h 04m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff315564f40946d68ce6173586f8210e : FAILURE in 1h 37m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0227c9e1ea5b4c2dac6892a99e1d5348 : FAILURE in 1h 43m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc0515827c2840b7b6a54e49962720da : SUCCESS in 1h 16m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/56ab0ad4dfc042cebb7d29e935523017 : SUCCESS in 1h 32m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df8a25b2bc284498834bdf6cbbbc3631 : SUCCESS in 1h 26m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9f5d2e0780c4a7d898141869841d12f : SUCCESS in 2h 11m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f09fdaa829c4bb6889bc7079ac40fb1 : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":3},{"id":"080cda523b870e1ea06080a5a9867d65f3b14ef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-19 20:53:42.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9e9300bd9f54c4a599a76d3bac2858191b8831a4","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-19 20:59:04.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/58/781758/4/check/dsvm-tempest-TOYOU-ACS5000-driver/61d4dd8 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"58070bb23aaad62bbd0b61592a45d41b6e81ed22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:49:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1bea7598bdf4df6936e4ae4f7f1909e : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22fb76042dec44349f90d10595ddbd5e : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6242f81551e46f78f68e0fabbab230d : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27cdc8f038484effb2fe547121cce5b6 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60de34a443b4b18bdc67c8538d546e8 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0af9752e3bb4defbef0d6503262ef06 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55fb32bf2f464e06aed67b612b185ecb : SUCCESS in 15m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/deab5012c53742ce9a749075e8c58ef7 : SUCCESS in 49m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e102e6282b154bb0ad93f470c6a2db15 : SUCCESS in 1h 22m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3f412429134b33bc5a60d52eea8be7 : SUCCESS in 6m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/edb261688a724b1f829358690ba22c0d : SUCCESS in 14m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8591fbf9108d47d1a9b2bac2df637f91 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d59f0d8f7004dcf9d603dd8087b253f : SUCCESS in 8m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42a34def38bf4d4f9a317058ff0efca9 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4df4adfec4c443aba5affe20fb07100d : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8415ce6ef5b244d1ae4b7de4a65ea045 : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efb0f04395574bd2bc040365c2f4b79d : SUCCESS in 1h 11m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b6f94d0f6bbc42f2be8e5b04e3f88169 : FAILURE in 1h 15m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b8b1ebe7d00493595d738a2ff3a7d1a : FAILURE in 1h 39m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/abe53249501d4eab863510f98f6b5c00 : SUCCESS in 1h 04m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b610f3bc36a4e329dd789b30c7a83cf : SUCCESS in 1h 29m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/15ced765d3a84715b2a7e226a811218b : FAILURE in 1h 33m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c89936166994947a521f66443f04e80 : SUCCESS in 1h 43m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7af3587541524cef9a65f0f5c3eea7fe : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":4},{"id":"b701b77a8f92ecf8394dbd681eb1879da2a525b0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-19 23:24:53.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781758/781758/4/check-not-vote/ext-cinder-zuul/fe2da8a : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":4},{"id":"dcee4bf6d24329896f4678ad0dd934d172ad20f0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-19 23:41:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/781758/4/ : SUCCESS in 1h 27m 07s","accounts_in_message":[],"_revision_number":4},{"id":"a681830caeea829b86ae1475748fd35fe919e44b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-20 01:24:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/58/781758/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e7224d5/ : FAILURE in 2h 14m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/58/781758/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d1874aa/ : POST_FAILURE in 1h 57m 55s","accounts_in_message":[],"_revision_number":4},{"id":"565be05d7ed47ecc9df6eecedd19c1131697a781","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-20 01:25:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/781758/4/check/dsvm-tempest-cisco-zonemanager-job/f2ed91e/ : FAILURE in 1h 22m 40s","accounts_in_message":[],"_revision_number":4},{"id":"28f4524dc40211d36d415ebbcbc18bf45efb91e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-20 03:37:54.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/781758/4/check/tempest-dsvm-full-focal-py3/5d9d116/ : SUCCESS in 2h 10m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/781758/4/check/devstack-plugin-ceph-dsvm-focal-py3/c0727d7/ : TIMED_OUT in 2h 42m 54s","accounts_in_message":[],"_revision_number":4},{"id":"6222e41f373b8a63297e219843ef1f2a87996f08","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-20 11:14:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/781758/4/check/dsvm-tempest-zadara-driver/7f03515 : SUCCESS in 1h 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7ef3140cfc1c848df83c903851225c52d84e87a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-21 05:42:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"09316ac22dd0d78c9f9a3c2fc00ed4a5c3421038","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-21 07:16:19.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781758-2/PVME-iSCSI : SUCCESS in 1:03:11\n* PVME-FC http://os-logs.tristero.net/dell-ci/781758-2/PVME-FC : SUCCESS in 55:53.03","accounts_in_message":[],"_revision_number":2},{"id":"a9ac78cb5e44ba181eae28aa2320886c20436762","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-21 16:52:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-1/ns5_nfs : SUCCESS in 4h 52m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-1/ns5_iscsi : SUCCESS in 4h 56m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d15543cb88bc7fb7f833a0c6e0204b87e837295c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-21 17:56:57.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-2/ns5_nfs : SUCCESS in 5h 15m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-2/ns5_iscsi : FAILURE in 53m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"b27cfc41e88cc532e4e5d62bcc2b8687efb3db45","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-21 20:08:41.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-3/ns5_nfs : SUCCESS in 5h 12m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-3/ns5_iscsi : SUCCESS in 5h 5m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"afa8c3207940c50dfd5d1319b4f43e1dab50573f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-21 22:11:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-4/ns5_nfs : SUCCESS in 5h 17m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-4/ns5_iscsi : SUCCESS in 5h 2m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"a53a8ab42aaef0ef5a0fdd18776ed54e9c70e993","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 15:24:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-4/ns5_nfs : SUCCESS in 4h 57m 10s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-781758-4/ns5_iscsi : SUCCESS in 4h 44m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6d176acd7a2ccdcc2f5c7255d3cc0ccf1318fd53","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-22 17:35:28.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1\n\nLooks like the CI problems are related to this: http://paste.openstack.org/show/803784/","accounts_in_message":[],"_revision_number":4},{"id":"f67d10d8fd89a6c575c3abd606020a372d972f16","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-22 18:43:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/58/781758/4/check/DellEMC_Unity_FC/0ed14b2/DellEMC_Unity_FC/935 : SUCCESS in 1h 24m 54s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/58/781758/4/check/DellEMC_Unity_iSCSI/8aa6c84/DellEMC_Unity_iSCSI/887 : FAILURE in 3m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bd6ff0d728bab193e9740a0d031a2f5779590956","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 03:53:55.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/781758/4/check/DellEMC_XtremIO_FC/a88e00b/DellEMC_XtremIO_FC/1098 : FAILURE in 2h 03m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/781758/4/check/DellEMC_XtremIO_iSCSI/0ea382b/DellEMC_XtremIO_iSCSI/1103 : SUCCESS in 1h 47m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"9cb85bb6060761aca632b01c5c8806e0908e05c1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 13:22:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/781758/4/20210323195953/NEC-ISCSI : SUCCESS in 2h 18m 36s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/781758/4/20210323195953/NEC-FC : SUCCESS in 2h 22m 22s","accounts_in_message":[],"_revision_number":4},{"id":"660cabb3e1b999e691dc73c41cbd6d4d56fa3851","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-27 16:55:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9028fd25ef2a98ba3939faaebce585c9f53b747","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-02 16:30:07.000000000","message":"Abandoned\n\nCleaning up.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a95789997e64894f2ca4846ca5e1881874be0c3c","revisions":{"b4b061ab3d368c54bdfa7e9522af2d8d389f7b0e":{"kind":"REWORK","_number":1,"created":"2021-03-19 14:58:21.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/58/781758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/781758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/781758/1"}}},"commit":{"parents":[{"commit":"0dd28a16c56d2c30929e0503038afc050435461c","subject":"Merge \"Changed whitelist to allowlist in tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dd28a16c56d2c30929e0503038afc050435461c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 14:52:21.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 14:58:11.000000000","tz":0},"subject":"WIP: Test NFS CI","message":"WIP: Test NFS CI\n\nDepends-On: http://review.opendev.org/c/openstack/cinder/+/749155/\nDepends-On:\nhttp: //review.opendev.org/c/openstack/devstack-plugin-nfs/+/742905/\nChange-Id: Ie3c6367c01aaec6e5f97ccb17343022c9308c7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4b061ab3d368c54bdfa7e9522af2d8d389f7b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4b061ab3d368c54bdfa7e9522af2d8d389f7b0e"}]},"branch":"refs/heads/master"},"49c866312c24c02021688f01623a9e5692a44d7f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-19 15:00:01.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/58/781758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/781758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/781758/2"}}},"commit":{"parents":[{"commit":"0dd28a16c56d2c30929e0503038afc050435461c","subject":"Merge \"Changed whitelist to allowlist in tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dd28a16c56d2c30929e0503038afc050435461c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 14:52:21.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 14:59:49.000000000","tz":0},"subject":"WIP: Test NFS CI","message":"WIP: Test NFS CI\n\nDepends-On: http://review.opendev.org/c/openstack/cinder/+/749155/\nDepends-On: http://review.opendev.org/c/openstack/devstack-plugin-nfs/+/742905/\nChange-Id: Ie3c6367c01aaec6e5f97ccb17343022c9308c7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49c866312c24c02021688f01623a9e5692a44d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49c866312c24c02021688f01623a9e5692a44d7f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ba8b06eb3ac1c62e330dc2dbf4329c7ed37c1bc3":{"kind":"REWORK","_number":3,"created":"2021-03-19 18:05:06.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/58/781758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/781758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/781758/3"}}},"commit":{"parents":[{"commit":"0dd28a16c56d2c30929e0503038afc050435461c","subject":"Merge \"Changed whitelist to allowlist in tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dd28a16c56d2c30929e0503038afc050435461c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 14:52:21.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 18:04:16.000000000","tz":0},"subject":"DNM Test NFS CI","message":"DNM Test NFS CI\n\nDepends-On: http://review.opendev.org/c/openstack/cinder/+/749155/\nChange-Id: Ie3c6367c01aaec6e5f97ccb17343022c9308c7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba8b06eb3ac1c62e330dc2dbf4329c7ed37c1bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba8b06eb3ac1c62e330dc2dbf4329c7ed37c1bc3"}]},"branch":"refs/heads/master"},"a95789997e64894f2ca4846ca5e1881874be0c3c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-19 20:53:42.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/58/781758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/781758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/781758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/781758/4"}}},"commit":{"parents":[{"commit":"0dd28a16c56d2c30929e0503038afc050435461c","subject":"Merge \"Changed whitelist to allowlist in tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dd28a16c56d2c30929e0503038afc050435461c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 14:52:21.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-19 20:53:30.000000000","tz":0},"subject":"DNM Test NFS CI","message":"DNM Test NFS CI\n\nDepends-On: http://review.opendev.org/c/openstack/cinder/+/749155/\nDepends-On: http://review.opendev.org/c/openstack/devstack-plugin-nfs/+/742905/\nChange-Id: Ie3c6367c01aaec6e5f97ccb17343022c9308c7e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a95789997e64894f2ca4846ca5e1881874be0c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a95789997e64894f2ca4846ca5e1881874be0c3c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
