)]}'
{"id":"openstack%2Fcinder~781788","triplet_id":"openstack%2Fcinder~master~I482c3d5520a7bbb9971942cdb4c4208052106d70","project":"openstack/cinder","branch":"master","topic":"bug/1920237","hashtags":[],"change_id":"I482c3d5520a7bbb9971942cdb4c4208052106d70","subject":"Backup manager: Synchronously call remove_export","status":"MERGED","created":"2021-03-19 16:13:57.000000000","updated":"2021-03-28 09:04:51.000000000","submitted":"2021-03-25 17:00:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781788-bug/1920237","meta_rev_id":"b82dda5307b9be22bf2f909c7dfee1c093630777","_number":781788,"virtual_id_number":781788,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 17:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-24 20:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-24 06:25:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-24 00:58:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2021-03-24 14:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-24 20:08:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 16:19:21.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-19 19:09:26.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 19:52:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 20:03:12.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-03-19 20:21:13.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 20:48:34.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-19 21:17:26.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-19 21:27:32.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-19 21:32:50.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 21:48:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-19 23:18:08.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 06:08:54.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 15:15:14.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-20 15:23:05.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-20 15:26:33.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-20 20:36:38.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-20 21:39:02.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 23:25:07.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 05:43:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 14:52:12.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-21 18:33:57.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-21 20:05:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 15:06:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-24 00:58:02.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-03-24 03:10:27.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 04:02:32.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 06:25:35.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-03-24 14:51:01.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-26 04:59:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-26 11:10:38.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d3d28d671b7013eba808b6e06ab3065c0859268e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-19 16:13:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef5142486dfed8a87fe1f841455c717d7e2a0fa7","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-19 16:19:21.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/88/781788/1/check/dsvm-tempest-TOYOU-ACS5000-driver/13b70af : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ffd6849cf82dd18364b49be704050875a70e2b0d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-19 16:24:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"75ee809bcee1ae47badb30fa5f922ffea4958247","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-19 19:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781788/781788/1/check-not-vote/ext-cinder-zuul/1c7cdcd : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c2af5859fe9c0c3bceda8fb88cfc71adef36909b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 19:52:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a827bec24404e4f8be3b3664c1d1d4d : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0616b31ebbd7444b9d98d9982ae7b52e : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04be51b053f34c749003285a32f4574c : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa05a5ecfbb041699017b8d9d86a77e9 : SUCCESS in 15m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ba5310c280c4db38f7464a12d2823c9 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aef5fdb219540159546b95194654665 : SUCCESS in 18m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/35d697170fbc4557b54af2bbe5ade4c0 : SUCCESS in 1h 06m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85f423414719470d9e07ca94b9ae0edd : SUCCESS in 1h 29m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e467887b26714a869332ec57ce68aab8 : SUCCESS in 23m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1099d8ed0ee4cc98c579b6d6b63b53d : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb932599c72c47d6891399f630b19759 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ef2f5501b984f7da68534770c3ee83d : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/27ad90313b294cd08a26df3c18328f0a : SUCCESS in 38m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a67ba99792074d6589683d3e27c75d98 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c5f6d48ab2854f279dc31f421cc3a9dc : SUCCESS in 1h 43m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/775e2a4c3e7d41888ca59731d10beaba : FAILURE in 1h 52m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc3324dcdac344a6bb013981753cbd10 : SUCCESS in 1h 34m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/748f5037682240f78d708b015c9235ca : SUCCESS in 1h 20m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f73c5675012c4b669f6a7e920a916f90 : FAILURE in 1h 24m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ba9951df5994ddcb14a496a660c6fd6 : SUCCESS in 1h 40m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a46cba14eb04db69639baef5cb6beeb : SUCCESS in 1h 57m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51c24ec6a8f54a8b83349a0b59c4a69e : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"80cf6532112e35e8bcd0c69a3874f747b07d1922","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-19 20:21:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c6a002d5c7a1e72a97aac515160382e47e1a1a36","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-19 20:48:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/88/781788/1/check/DellEMC_PowerStore_iSCSI/edd821a/DellEMC_PowerStore_iSCSI/657 : SUCCESS in 55m 31s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/88/781788/1/check/DellEMC_PowerStore_FC/2532d08/DellEMC_PowerStore_FC/651 : SUCCESS in 50m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"636c5a1f95654f6a7dff1daa1bac6f074c3059ae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-19 21:17:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/88/781788/1/check/dsvm-tempest-cisco-zonemanager-job/62b3c18/ : FAILURE in 1h 21m 17s","accounts_in_message":[],"_revision_number":1},{"id":"386c4acf018e47e1f05708270b04a9beaf7a13b9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-19 21:27:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/781788/1/upstream-solidfire-check/cinder-solidfire-iSCSI/a70ee20/ : FAILURE in 1h 33m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"348e774e6a999c6702b7d8165b9d5b49356455f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-19 21:32:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/781788/1/upstream-check/cinder-cDOT-iSCSI/330cf7b/ : SUCCESS in 39m 51s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/781788/1/upstream-check/cinder-cDOT-NFS/073ab73/ : SUCCESS in 37m 11s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/781788/1/upstream-check/cinder-cDOT-FCP/10f956f/ : FAILURE in 1h 38m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"058ee19a36dfe025094ec2d241811d8507566f15","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-19 21:48:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/88/781788/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/af1f31a/ : POST_FAILURE in 19m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/88/781788/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5fa6856/ : TIMED_OUT in 3h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"480f6577dae58c54ef0bb7be1afc85f50559ef7f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-19 23:18:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/88/781788/1/check/tempest-dsvm-full-focal-py3/a2b9708/ : SUCCESS in 1h 58m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/88/781788/1/check/devstack-plugin-ceph-dsvm-focal-py3/09e2345/ : SUCCESS in 2h 09m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6128e4a6af467de8e2432a4bd7da0ac28a31bc80","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-20 06:08:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/88/781788/1/check/dsvm-tempest-zadara-driver/d15b307 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d73418bb3ac2c5f8ecf46c2401f2181aab652b0b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 06:10:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"204a71ee3905d77c9affe81b61bf36de86a92e30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-20 11:12:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a5b80ee5db07999802a95fb16215ffd7e1e11287","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-20 11:12:15.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"12b7db363a9870f9185810c0d1f9f9f09cb38598","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 11:13:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a252017f9c88ff35c6e3932cca9c7de529baa87f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-20 11:17:32.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/88/781788/2/check/dsvm-tempest-TOYOU-ACS5000-driver/4770183 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"5538eb069e24250dc066bcecf244686794c634a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 11:31:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9b29fa0580e49c28ece8c60cfbc0df5 : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5b0d6cb3d0940ca945b39536fd5f9a2 : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c9443d0d856b324d3c9efe3f5621849c7094b12","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-20 11:52:57.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781788/781788/2/check-not-vote/ext-cinder-zuul/28bf0a2 : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3299fdd4ade5b5528fa7e54cecfa6d05ddda3dcf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-20 12:44:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/88/781788/2/check/dsvm-tempest-cisco-zonemanager-job/7ee9d60/ : FAILURE in 1h 21m 49s","accounts_in_message":[],"_revision_number":2},{"id":"152f6c939e3554d3c114a94caa6241ec519f0d6d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-20 13:12:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/88/781788/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c897a00/ : SUCCESS in 1h 59m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/88/781788/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/954496a/ : POST_FAILURE in 1h 56m 13s","accounts_in_message":[],"_revision_number":2},{"id":"74e8a5d9b2233cb18f2c2497826d14bf169a32c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 13:16:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7ec19bc3bdd419fa86df7471c59ec3f : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354f5e41ce984a2ab596ffe15399c93a : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c03e2d02556849e48f24b6804af46d67 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6c0d0159d454ea8a7312d5d4d33d718 : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0d949aa15b4ed3bb5a126725b512cb : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e419d6843ad456e85df35cdeb598360 : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c1db459f9f494d5f9647c8c82c4ff53d : SUCCESS in 57m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5faa0ceec01a4486b9af9246ab5cde8f : SUCCESS in 1h 27m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/70ab5cd0110b4408a0ca796a30acec3d : SUCCESS in 12m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab31ea9f1c4f4767ad122bb22d3a145c : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9949b024fe848cb9e9fc03519b08f11 : SUCCESS in 7m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e3c33eb80acb411780bb1ba361436f6c : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb22f3bef1c8470e850a0e56eda211d4 : SUCCESS in 36m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2fdfdb86a8324be7b1d612bdf9c95fd9 : SUCCESS in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57f7a26e59f346edaf4fb2b49efd53b5 : SUCCESS in 1h 33m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df74fd1c62cc460f827581ff691f419e : FAILURE in 1h 31m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30e839dd2805435da08c073bc914778f : SUCCESS in 1h 42m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/65f2f8f4db7f447bada9ad5a917ac47b : SUCCESS in 1h 26m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b428272ee304ab090346d5afa058d53 : SUCCESS in 1h 37m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4c7f49adf914d1ba1e79a13b85f6a72 : SUCCESS in 1h 36m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/571837c1d7554769bcb87f36fdf9e00f : SUCCESS in 1h 59m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba430bb7078943a6ad0da3c427da884c : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9a652dc60ea5eb150db32e9fc00b79d532019998","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-20 13:20:49.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e3ad9d5552564885a2d331b3ccb9c4f7 : POST_FAILURE in 1h 48m 16s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/54fa551e13ee4d86baa9af63518d1f51 : POST_FAILURE in 2h 00m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6cf34bc1243cc394665339fa2c7b16e3f89daf46","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-20 14:14:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/88/781788/2/check/DellEMC_PowerStore_iSCSI/49452b8/DellEMC_PowerStore_iSCSI/661 : SUCCESS in 57m 45s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/88/781788/2/check/DellEMC_PowerStore_FC/2c0be96/DellEMC_PowerStore_FC/655 : SUCCESS in 52m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"ff79845d84e983d6c53f6c942f5e0e36a0858f91","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-20 14:18:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/781788/2/upstream-solidfire-check/cinder-solidfire-iSCSI/50ad00f/ : FAILURE in 59m 59s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"15ca52b917d8f944b5ce9aa33b008a9222572ee2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-20 14:31:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/781788/2/upstream-check/cinder-cDOT-iSCSI/c1feeb6/ : SUCCESS in 42m 16s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/781788/2/upstream-check/cinder-cDOT-NFS/1fbc31f/ : SUCCESS in 39m 23s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/781788/2/upstream-check/cinder-cDOT-FCP/49e9854/ : SUCCESS in 1h 13m 26s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b6ccd8c5303f5fc975298f437053e92800a9aa25","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-20 15:15:14.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781788-2/Seagate-iSCSI : SUCCESS in 46:42.98\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781788-2/Seagate-FC : SUCCESS in 41:02.75","accounts_in_message":[],"_revision_number":2},{"id":"c30c89e8709614454b11ed1544df52d4171baa9f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-20 15:23:05.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781788-2/HPMSA-iSCSI : FAILURE in 1:02:01\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781788-2/HPMSA-FC : SUCCESS in 33:59.46","accounts_in_message":[],"_revision_number":2},{"id":"f456172b9a802ae566faf2a309981aecba5fea7c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-20 15:26:33.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781788-2/Lenovo-iSCSI : SUCCESS in 53:10.90\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781788-2/Lenovo-FC : SUCCESS in 46:07.22","accounts_in_message":[],"_revision_number":2},{"id":"7b635c1517f62eecd43454a14615b4c1e0b615b3","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-20 20:36:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/88/781788/2/check/DellEMC_SC_FC/c8d94b1/DellEMC_SC_FC/474 : SUCCESS in 1h 53m 57s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/88/781788/2/check/DellEMC_SC_iSCSI/0da9d75/DellEMC_SC_iSCSI/471 : SUCCESS in 2h 07m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"95bec83376f055480f27956290af12f85fea32a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 20:38:10.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"0b2a93b3d5ccefdb48292ad0f0fadefdcf64d824","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-20 21:39:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/88/781788/2/ : RETRY_LIMIT in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"d043412048d4789c31424caf957a649626124834","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 21:59:02.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"5d9c4c187216c6884c00e404532adf11a7f62fcb","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-20 22:44:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/88/781788/2/check/dsvm-tempest-zadara-driver/a3ea61c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18a6b9a28e5b2f37c35674c72d06dd20e5d0fca2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-20 23:25:07.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/88/781788/2/check/DellEMC_VNX_iSCSI/6416947/DellEMC_VNX_iSCSI/606 : SUCCESS in 3h 05m 43s\n- DellEMC_VNX_FC http://publiclogs.emc.com/88/781788/2/check/DellEMC_VNX_FC/688b89a/DellEMC_VNX_FC/645 : SUCCESS in 1h 27m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ac990967c90f62fd04fbbfbc0a1a8276e86ebda8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-21 02:53:22.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"45b41a3d31f2585490c616a883a778d9d65c6fe3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-21 05:43:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"371da66a8db8c17d2438ca698d7f9e0b1151bb80","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-21 14:52:12.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781788-1/PVME-iSCSI : SUCCESS in 1:01:54\n* PVME-FC http://os-logs.tristero.net/dell-ci/781788-1/PVME-FC : SUCCESS in 56:48.98","accounts_in_message":[],"_revision_number":1},{"id":"80749415fa09203b7c2baf7bd6c390b9938d7393","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-21 18:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-1/ns5_nfs : SUCCESS in 5h 7m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-1/ns5_iscsi : SUCCESS in 5h 2m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4b436b3c480f26f78c4c1e6eced3e4a8bf68d548","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-21 20:05:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/88/781788/2/check/DellEMC_PowerFlex/f58df06/DellEMC_PowerFlex/623 : SUCCESS in 2h 41m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7d104467c0ec133dbca745c257114a8cf12fb747","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-21 22:31:18.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781788-2/PVME-iSCSI : SUCCESS in 1:03:06\n* PVME-FC http://os-logs.tristero.net/dell-ci/781788-2/PVME-FC : SUCCESS in 56:35.33","accounts_in_message":[],"_revision_number":2},{"id":"90a966fe19e34c8480d96a6c499620a44aedb13e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-21 22:46:56.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b83101b84a5175f1b849f8de0828c3f1abb5a215","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 04:59:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-2/ns5_nfs : SUCCESS in 5h 2m 27s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-2/ns5_iscsi : SUCCESS in 5h 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"33da3fcd04b635b43b235738dd2a02cb6c949f27","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 07:19:54.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"503cf80f28d4a22ae373580522e7695c856ab08f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-22 14:38:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"edcb646f614dbc74767adccfb29c89ca6a89c392","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-22 14:44:22.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/88/781788/3/check/dsvm-tempest-TOYOU-ACS5000-driver/2ff3884 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"fd6a1fc5fe5145431a7931a630e99fec0067b3a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:01:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/718d49bac75f45d58a04585dbb8b72db : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/185db4545e8f432bb935ee909b355b8c : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"854b4af699f92f9b5173498fa11ca8371c81dfa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-22 15:06:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ebdd1d872329c3c8e664cba566a72ca528d851a3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-22 15:12:21.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/88/781788/4/check/dsvm-tempest-TOYOU-ACS5000-driver/e59440d : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"2fa9716252db0c7338ec667ef0181f2edc227bba","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-22 15:49:20.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781788/781788/4/check-not-vote/ext-cinder-zuul/0965db8 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":4},{"id":"118b178a3c01d5b0be060b21cf39406e411d5814","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 17:34:42.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"f58c821a0c36788e63ad3a87cc92ebce52873163","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-22 18:26:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/88/781788/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4dff611/ : SUCCESS in 2h 15m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/88/781788/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e6c5a32/ : SUCCESS in 2h 06m 48s","accounts_in_message":[],"_revision_number":4},{"id":"78c0c9eadaad50b666e1a4bc6c64e417ce764394","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-22 19:00:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/88/781788/4/check/dsvm-tempest-cisco-zonemanager-job/62ef10b/ : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":4},{"id":"463e97f0fac4f412b59294c4f6c5bd08cadccdd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 19:34:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f346df47460740b6829ff27881c3b1dc : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256a324b4a2248c8b231cfef8b8ca2fe : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6537529b728945488500d397e315add2 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6026fea00619429e82bf796a8f664a45 : SUCCESS in 10m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1b45cfb86e447f99d5ef929ebd3e89f : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4729191b53114ce48ed794e9b0ba40b0 : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/df49ecb371b8476b8cbcb8b89d3a7b1e : SUCCESS in 1h 05m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a137b2878c484f89a69fc52ca61ed094 : SUCCESS in 1h 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb17bd2437d14683b082a59f62244bac : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7e0bf617849a4e9287b7546118519a28 : SUCCESS in 16m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00e66c7aa1b54bcd8c6595404b6bd1e7 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25851dd8b9d047eba4a57c6a28498a8c : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8981787ead1c40289f3b85082ba2f31b : SUCCESS in 7m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebd70f70e94545b786ddd27641bcde50 : SUCCESS in 1h 13m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/016d630229124174a55750176f8bf7a1 : SUCCESS in 8m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/417e4d25512f4c828b7f5310d029a142 : SUCCESS in 1h 05m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa280627fa2a470bb871a94581d10b7e : FAILURE in 2h 03m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4ebddd353384bd8818ae0076912be16 : SUCCESS in 1h 49m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ec9d1e25257846be9ee5899907f9c9c0 : SUCCESS in 1h 18m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cb7b9ca929ff476bafd2d017e2254340 : SUCCESS in 1h 32m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f0ac5e19fe274b5dbfbec0f109bb25a2 : SUCCESS in 1h 54m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd07c7582dd044f3a2bc2634c99691b9 : SUCCESS in 1h 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/962fb0a4d1d84a9494836ea00bad8b52 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9c1ce29ec1c6518dbcc28db3fd8f62974db87401","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-22 20:17:38.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerStore_iSCSI/b42cfa3/DellEMC_PowerStore_iSCSI/673 : FAILURE in 39m 04s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerStore_FC/5afe402/DellEMC_PowerStore_FC/667 : FAILURE in 42m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"e756cfd3dff03019d0e5db3b317ea43ba24326ba","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-22 20:17:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/781788/4/upstream-check/cinder-cDOT-iSCSI/e64aca2/ : FAILURE in 2m 47s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/781788/4/upstream-check/cinder-cDOT-NFS/012b08a/ : SUCCESS in 42m 22s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/781788/4/upstream-check/cinder-cDOT-FCP/04e049a/ : FAILURE in 2m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8371ceb36ecdb8c1cda912967a38d1214fe692bd","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-22 21:29:16.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/781788-4/Seagate-iSCSI : SUCCESS in 44:02.88\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/781788-4/Seagate-FC : SUCCESS in 38:13.10","accounts_in_message":[],"_revision_number":4},{"id":"f58ed6ba944e133eb1b190cc0942e340bb4eba2f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-22 21:37:16.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/781788-4/Lenovo-iSCSI : SUCCESS in 46:59.60\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/781788-4/Lenovo-FC : SUCCESS in 43:25.35","accounts_in_message":[],"_revision_number":4},{"id":"a14b844fe38a38511a3439beda372a1689de6ca0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-22 21:40:39.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/781788-4/HPMSA-iSCSI : FAILURE in 58:01.05\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/781788-4/HPMSA-FC : SUCCESS in 34:32.38","accounts_in_message":[],"_revision_number":4},{"id":"ac3b3f59ecac58b68c7034e3e84b0339700b3805","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-22 21:43:49.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/88/781788/4/check/tempest-dsvm-full-focal-py3/bb3eeee/ : SUCCESS in 2h 15m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/88/781788/4/check/devstack-plugin-ceph-dsvm-focal-py3/e2ebd5a/ : SUCCESS in 2h 32m 58s","accounts_in_message":[],"_revision_number":4},{"id":"7532efd15dcde7318f1cbbdecd501f9a10abd0f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-22 22:02:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/88/781788/4/check/iscsi-driver-nimble/d0f7674/d0f7674.html : SUCCESS in 2h 01m 49s","accounts_in_message":[],"_revision_number":4},{"id":"131fc1f0e14461f3b8623f2f405b6c5a5e5eba83","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-22 22:04:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/781788/4/upstream-solidfire-check/cinder-solidfire-iSCSI/3b2161d/ : FAILURE in 2h 28m 19s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"c5ca44b2b9800371fc36e996547b06f623a99935","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-22 22:29:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/88/781788/4/check/dsvm-tempest-zadara-driver/8f52357 : SUCCESS in 1h 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9f4160c0a5fbc2d45c9c8d6362d676130bcaceb1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-22 23:16:02.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_SC_FC/d8d5652/DellEMC_SC_FC/486 : FAILURE in 1h 56m 24s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_SC_iSCSI/954434f/DellEMC_SC_iSCSI/483 : FAILURE in 2h 08m 31s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"e3cfc88801c83d55180951e321544abc376309ac","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 23:21:13.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-3/ns5_nfs : SUCCESS in 5h 3m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-3/ns5_iscsi : SUCCESS in 4h 53m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a0dba464a81e6c4c27f5c2319ae09649a4813f6f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 23:30:01.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-4/ns5_nfs : SUCCESS in 4h 54m 2s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-88-781788-4/ns5_iscsi : SUCCESS in 4h 57m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"8cf1feb7dfa6795880cf0f1e116f29f8285669c3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 01:12:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerFlex/c68d0c5/DellEMC_PowerFlex/632 : FAILURE in 2h 53m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b0020f47ce66f3f352642ab36353c1320b718ac3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 03:06:31.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_VNX_iSCSI/f10f275/DellEMC_VNX_iSCSI/616 : FAILURE in 3h 38m 14s\n- DellEMC_VNX_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_VNX_FC/5ac4c32/DellEMC_VNX_FC/656 : FAILURE in 4m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"5b246f8a133a111f7a18be9ade53f31065450273","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 06:00:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d7102a9c431a2a3df5e7668e3886532db57a323e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-24 00:58:02.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM and CI passed. \nLooks like no unittest is needed. \n```\n{6} cinder.tests.unit.targets.test_base_iscsi_driver.TestBaseISCSITargetDriver.test_remove_export_show_error [0.046036s] ... ok\n\n```","accounts_in_message":[],"_revision_number":4},{"id":"95fac990e87072c5d29d4a4f4bc64f16e8f3d8b4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-24 03:10:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerMAX_FC/42eb50f/DellEMC_PowerMAX_FC/610 : SUCCESS in 5h 53m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerMAX_iSCSI/9598ba6/DellEMC_PowerMAX_iSCSI/579 : FAILURE in 4h 57m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"6555d7d072dc60f70542e6a3c7bccb0172fd7e39","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-24 04:02:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/781788/4/20210324104138/NEC-ISCSI : SUCCESS in 2h 18m 48s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/781788/4/20210324104138/NEC-FC : SUCCESS in 2h 20m 51s","accounts_in_message":[],"_revision_number":4},{"id":"419ac01b55bafb43c1ce73aa7e33e3fef4686b5e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-24 05:40:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7870642bba29799fe50ccd888228fa2574083ff0","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-24 06:25:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3860b36e6b5357ad3a74877fc617d72cba3984ce","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-24 14:51:01.000000000","message":"Patch Set 4: Code-Review+2\n\nIf this helps the check it would be great.  Thanks for making this change.","accounts_in_message":[],"_revision_number":4},{"id":"39c368c68c97c8fdf3ae1b9a607e1f4f3da9d886","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-24 20:08:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM.  Thanks, Eric.","accounts_in_message":[],"_revision_number":4},{"id":"a4a3ccaa4741a37bab4b216b783c60581eac3fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 20:09:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0e97df54d07f0cb070330722e6cbc740ee6d8d8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:56:11.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0ce4102c2d74770aed1acc1d138a0c5 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651c5aa003c84d449cf6b3622fc05abe : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/858386c01aa34cddad7a78d61085c73e : SUCCESS in 9m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2927c59affca4be3acba6ab1ef44e221 : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d97ff7d586964ff89192756019cc24da : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/526f4c16f2364cfc8ca7b4f2355ece42 : SUCCESS in 1h 05m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/441870d47236404f9217919747b592cc : SUCCESS in 1h 21m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cdb89577e7a45dbbba770e65b95d9c8 : SUCCESS in 7m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b5aa314568d47169028089cf588f320 : SUCCESS in 1h 11m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3fdcbf91fbf7474e84bc24634d6f11e4 : SUCCESS in 1h 11m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebea69aedb2f46ce930b1aa8587edb0a : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":4},{"id":"fdab3fbdd5efa78ca3c22e01a0dc9af83f6aa91a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-25 09:00:57.000000000","message":"Patch Set 4:\n\nAdding unit tests and doing the same to the remove snapshot export would have been good.","accounts_in_message":[],"_revision_number":4},{"id":"bdb1cca8f22d3f7aa30dff16ea58d6f236f4f87f","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-25 09:01:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bb388cc7f2f0852efffb0f57123f4e2a4ec3d68d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-25 09:03:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"ffa39a301bf803a7bb7198e37339a0615b5c3a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 09:25:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/944fc3883ed44db4b7fdb75fd52e1aa2 : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/246398fa7cff40afb8e3e2e0f89b2ea0 : SUCCESS in 18m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dfc8de14d1587de5a00ceac22498fb6c5a6d1e56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-25 10:16:03.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"782f4518e9bd2fc1b90064d19ec1d59e9c0c3ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:03:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/614a767dffac4be1afc605d16e015133 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee6491178e9441d93ce8760a3f13873 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2526efd37c84db98e47f9c44be9b8f8 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df202e84152420da412a5bf8181980f : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cec836d99554c6684e180ef66e55f6b : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7988ad7fd870443d82437b304e287f8f : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2d43968117e743328c8e48f4dcf1fbfd : SUCCESS in 55m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fc62756dddd4342bfdfba07111ccb2b : FAILURE in 1h 04m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e5603ba356a4b529e569d3712bcbae8 : SUCCESS in 6m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3bed50272ac84415b7b92217da37ca37 : SUCCESS in 13m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c4c2abbd7b540dba7d8ba37811db071 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b9674acc5f441ebb7d8e6f1d94d6d10 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec65985dc7b04a92bed11333246f09da : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8dc00bfd53524c00be709ce3d43215ab : SUCCESS in 34m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/202ff4f11311408b8f78d1a5e2d6e0c2 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb9425b4d4844c89a204b39d944fb540 : SUCCESS in 1h 10m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd603a40314541bcb06a56c2e480ddd2 : FAILURE in 1h 27m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01f53c2956c74b59a4e15ab301073dff : SUCCESS in 1h 34m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/17fc74e31dd64ba88c61e4c7fdf46610 : SUCCESS in 1h 01m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9b26b3dbf2224e0493cb9854cf263e80 : SUCCESS in 1h 07m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77e3aec285774a76b2586697380f3825 : FAILURE in 1h 35m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a57495a64ec4be382c733c768f4c619 : SUCCESS in 1h 59m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d0d4e64ddb1479e8459e29651f14c2d : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":4},{"id":"fcc3bb304c5b4004060f4ead5190ea68ffe8a4c2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-25 11:40:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/88/781788/4/check/dsvm-tempest-cisco-zonemanager-job/dba38cd/ : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":4},{"id":"9bde892b78af6714e9c8c576c5dc658d4e6188b7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-25 12:17:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"10ca551cec0b57eee561c45dbc2fffbed790e52d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-25 12:18:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"117d10a903feeec3691ea43a377e6cec0be1bf33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 12:54:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a10d7c796fb449f186c8bbd04ef44be5 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a335537bfe49460591f37c2d1a72a248 : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"afcf7bdd99013e101e6c031da81cfc64ef4fcf56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-25 13:36:33.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"e8fdbb17d8f2a8eb9cc78040f5b57c8f75102f4c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-25 13:45:55.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/88/781788/4/check/dsvm-tempest-cisco-zonemanager-job/7b4b606/ : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":4},{"id":"0b4ea1050fd6b6ca4e54af97dbfee78bd6e507d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:39:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d9351b0ad174ef786846a6866f6a1a8 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ddb1e9de4514eae80c2fd0d8a345f1f : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7b2b0c590ca4fd0a289ee0da69ac6f1 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eece650fe624ebb8cf925985e256519 : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54f045d8459f4bcab34dd7ca6560d98b : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/893f494c3b9942669585c031f43a86f7 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/f210c56750644fe28865921477d7244f : SUCCESS in 1h 03m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5f84045c5f54e3289a2a4590db18a9c : SUCCESS in 1h 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b123dbf74842ad8806e575aa4ac23d : SUCCESS in 7m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd54202304d448debcf2b2bf609d698e : SUCCESS in 13m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3d181deb4704c6d98722a1edf37f488 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c55631c90694984bd1c3637b6652f36 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5c025ada8b641bebe513dde5c909337 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/efb6b9f21bee4bc3844b693e38cad32e : SUCCESS in 48m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff7f0079d2e449f09dfdfa783daa0519 : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d956e6cdbe484dfcbd4fa4d611945bf0 : SUCCESS in 1h 18m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7d3edae6042145b18e8eb95331539847 : FAILURE in 1h 27m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37a9e5a66ad84e9b9b6f7c24d8328a79 : SUCCESS in 1h 31m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5ca44cbbb324b74a9bd8609185bb325 : SUCCESS in 1h 16m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/12da71c0cede4f939d92eeb08de8ed96 : SUCCESS in 1h 40m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e45d21a90e840cb96fc7209b8f8a9fa : SUCCESS in 1h 18m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8724eff300e74261b2734e74fc54fa4f : SUCCESS in 2h 03m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ab298647caa45f7b900f3148924d2b0 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":4},{"id":"dcbef9b30f29a43e0877ba5e5c6aaa85cc8ad550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:40:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a6b7e070c1f839b3596ffcf7365a6a8bfc05a39","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-25 15:31:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerStore_iSCSI/73a58dd/DellEMC_PowerStore_iSCSI/704 : FAILURE in 36m 06s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerStore_FC/b4a45ca/DellEMC_PowerStore_FC/698 : FAILURE in 36m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"ee6a1c165285a6607087b5aea115031ff6e3b543","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-25 15:49:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/781788/4/upstream-check/cinder-cDOT-iSCSI/ccd3732/ : SUCCESS in 43m 48s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/781788/4/upstream-check/cinder-cDOT-NFS/45f6241/ : SUCCESS in 40m 54s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/781788/4/upstream-check/cinder-cDOT-FCP/6e523a2/ : SUCCESS in 1h 08m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"212747a6f4975691bb99fade63a4c273532aefe6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-25 16:23:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/781788/4/upstream-solidfire-check/cinder-solidfire-iSCSI/cb040c5/ : FAILURE in 1h 42m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"c3ba4c6b80165c77967037a28d16943f464b91dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-25 16:51:07.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"e5a8e4ab1af0213595e766d0faa60dacc6a81832","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:00:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8762bc3d35cb46299afed24ff3beaae2 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbea6440de04bbbb9bb0f3d9c33098b : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ca50c2cb7014386af2d635b3551b32f : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c7fd4c29c9a4b9bbf4b5c36e0f93f7e : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1bd3675e804b598098e0e0b928da29 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1238c74217a246569427f42959bbc5e8 : SUCCESS in 1h 20m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/865258d8cf2544df86c31adaf146bb66 : SUCCESS in 1h 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb3b9d0b6c1143198152cfdea8bedbf5 : SUCCESS in 6m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a66e92e16e26412ba6bfa89c95034cb6 : SUCCESS in 1h 17m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/56847812aa8148f3aab7c591cb20ff54 : SUCCESS in 1h 22m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa3c04656c964bc690bae46c6f9f4618 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":4},{"id":"671cc8a3336dafc8373148aaa3dcb41716b03170","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:00:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"563cfb0157c6a7f7d307047fee3e33626e9d78b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:05:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ddc1b8a5a8432a8ae68a6294463183 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d8c4767f402409aaad7b1efe230aa36 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4},{"id":"15a4e0142ccbae407af61df976ab8aac8d390107","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-25 19:06:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/88/781788/4/check/DellEMC_PowerFlex/6d69ae8/DellEMC_PowerFlex/657 : FAILURE in 7m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e6d2ee8d02c4f0db0e220f7ddbfcba3e7c809d0","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-25 19:42:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/88/781788/4/check/dsvm-tempest-zadara-driver/be1cf3f : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75db88dff9316099fa6e85dd7973de4eb1e9fbda","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-25 21:00:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_SC_FC/069b933/DellEMC_SC_FC/503 : SUCCESS in 2h 00m 43s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_SC_iSCSI/b9acd50/DellEMC_SC_iSCSI/500 : FAILURE in 2h 10m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"1ddd4a67a43296e2490a419e1fd4ad8a01f04869","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-25 21:01:51.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"5a2cb9a92dab74914b5d9dcf84aa60b2734a170c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-25 22:21:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"766ce3af5bcbe7db263bf960fd5aacb2b824c295","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 02:56:34.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_VNX_iSCSI/a71399b/DellEMC_VNX_iSCSI/638 : SUCCESS in 3h 33m 19s\n- DellEMC_VNX_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_VNX_FC/3a42e26/DellEMC_VNX_FC/684 : SUCCESS in 1h 11m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"6a66a268542e2fa5e2cb0e886139f11379a5d023","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-26 04:59:44.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_Unity_FC/0cf926d/DellEMC_Unity_FC/979 : SUCCESS in 1h 25m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_Unity_iSCSI/6e350b8/DellEMC_Unity_iSCSI/921 : SUCCESS in 1h 49m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6145c9ea3f3849a1a5ae6a1f9605adebf6e5dce4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 09:17:11.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"785eb5189c946d0199aa24833acb8900768997c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-26 10:46:01.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/88/781788/4/check/tempest-dsvm-full-focal-py3/0401eb3/ : SUCCESS in 1h 58m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/88/781788/4/check/devstack-plugin-ceph-dsvm-focal-py3/ccaf4a0/ : SUCCESS in 2h 25m 30s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud finger://zuul-v3-executor.novalocal/995d024747ab46abb3a2cd652aab91f3 : TIMED_OUT in 2h 39m 01s","accounts_in_message":[],"_revision_number":4},{"id":"a36b8cf061f19737c05f91aa5f621c69711b9047","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 11:10:38.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/88/781788/4/check/DellEMC_XtremIO_FC/26658d9/DellEMC_XtremIO_FC/1135 : SUCCESS in 2h 02m 57s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/88/781788/4/check/DellEMC_XtremIO_iSCSI/65e241f/DellEMC_XtremIO_iSCSI/1153 : SUCCESS in 1h 45m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"7521fd5be6389d9814d189b90dd291d3adf8afd2","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-26 14:39:45.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/781788-4/PVME-iSCSI : FAILURE in 1:42:05\n* PVME-FC http://os-logs.tristero.net/dell-ci/781788-4/PVME-FC : SUCCESS in 56:53.63","accounts_in_message":[],"_revision_number":4},{"id":"7101bd8592645a7b3d5b1abc4e84ade71037f6fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 15:22:50.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/781788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"b82dda5307b9be22bf2f909c7dfee1c093630777","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-28 09:04:51.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/781788/781788/4/check-not-vote/ext-cinder-zuul/7c536ea : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d6895f9d2aab84c116efc2f853bb5f9e57fac539","revisions":{"6c8e2c11c1aac46449e6f739d8e37173c5caf58f":{"kind":"REWORK","_number":1,"created":"2021-03-19 16:13:57.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/781788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/781788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/781788/1"}}},"commit":{"parents":[{"commit":"0dd28a16c56d2c30929e0503038afc050435461c","subject":"Merge \"Changed whitelist to allowlist in tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dd28a16c56d2c30929e0503038afc050435461c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 16:01:33.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 16:10:39.000000000","tz":0},"subject":"Backup manager: Synchronously call remove_export","message":"Backup manager: Synchronously call remove_export\n\nCall the volume\u0027s rpcapi remove_export method with\na \"call\" instead of a \"cast\" when detaching volumes\nin the backup manager.\n\nThis prevents _detach_volume() from returning before\nthe remove_export() has completed, which helps prevent\nproblems where remove_export() races against a\nsubsequent operation.\n\nCloses-Bug: #1920237\n\nChange-Id: I482c3d5520a7bbb9971942cdb4c4208052106d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c8e2c11c1aac46449e6f739d8e37173c5caf58f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c8e2c11c1aac46449e6f739d8e37173c5caf58f"}]},"branch":"refs/heads/master"},"47eda4913d3a72bd54cef5f1301fa7f97cf59da2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-20 11:12:09.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/781788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/781788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/781788/2"}}},"commit":{"parents":[{"commit":"2462abec42bbaceefcc9af0d1d0658e4f9b691f8","subject":"Merge \"add openstack-python3-wallaby-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2462abec42bbaceefcc9af0d1d0658e4f9b691f8"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 16:01:33.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-20 11:12:09.000000000","tz":0},"subject":"Backup manager: Synchronously call remove_export","message":"Backup manager: Synchronously call remove_export\n\nCall the volume\u0027s rpcapi remove_export method with\na \"call\" instead of a \"cast\" when detaching volumes\nin the backup manager.\n\nThis prevents _detach_volume() from returning before\nthe remove_export() has completed, which helps prevent\nproblems where remove_export() races against a\nsubsequent operation.\n\nCloses-Bug: #1920237\n\nChange-Id: I482c3d5520a7bbb9971942cdb4c4208052106d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47eda4913d3a72bd54cef5f1301fa7f97cf59da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47eda4913d3a72bd54cef5f1301fa7f97cf59da2"}]},"branch":"refs/heads/master","description":"Rebase"},"a37eaabf3dfce78c05a542553258d70e01b32bb8":{"kind":"REWORK","_number":3,"created":"2021-03-22 14:38:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/781788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/781788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/781788/3"}}},"commit":{"parents":[{"commit":"0dd28a16c56d2c30929e0503038afc050435461c","subject":"Merge \"Changed whitelist to allowlist in tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dd28a16c56d2c30929e0503038afc050435461c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 16:01:33.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-22 14:15:15.000000000","tz":0},"subject":"Backup manager: Synchronously call remove_export","message":"Backup manager: Synchronously call remove_export\n\nCall the volume\u0027s rpcapi remove_export method with\na \"call\" instead of a \"cast\" when detaching volumes\nin the backup manager.\n\nThis prevents _detach_volume() from returning before\nthe remove_export() has completed, which helps prevent\nproblems where remove_export() races against a\nsubsequent operation.\n\nCloses-Bug: #1920237\n\nChange-Id: I482c3d5520a7bbb9971942cdb4c4208052106d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a37eaabf3dfce78c05a542553258d70e01b32bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a37eaabf3dfce78c05a542553258d70e01b32bb8"}]},"branch":"refs/heads/master"},"d6895f9d2aab84c116efc2f853bb5f9e57fac539":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-22 15:06:29.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/88/781788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/781788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/781788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/781788/4"}}},"commit":{"parents":[{"commit":"2462abec42bbaceefcc9af0d1d0658e4f9b691f8","subject":"Merge \"add openstack-python3-wallaby-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2462abec42bbaceefcc9af0d1d0658e4f9b691f8"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 16:01:33.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-03-22 15:04:16.000000000","tz":-240},"subject":"Backup manager: Synchronously call remove_export","message":"Backup manager: Synchronously call remove_export\n\nCall the volume\u0027s rpcapi remove_export method with\na \"call\" instead of a \"cast\" when detaching volumes\nin the backup manager.\n\nThis prevents _detach_volume() from returning before\nthe remove_export() has completed, which helps prevent\nproblems where remove_export() races against a\nsubsequent operation.\n\nCloses-Bug: #1920237\n\nChange-Id: I482c3d5520a7bbb9971942cdb4c4208052106d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6895f9d2aab84c116efc2f853bb5f9e57fac539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6895f9d2aab84c116efc2f853bb5f9e57fac539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
