)]}'
{"id":"openstack%2Fcinder~782405","triplet_id":"openstack%2Fcinder~master~I336a8c92bfdd852cab42c9e5f992a30f3632bfaf","project":"openstack/cinder","branch":"master","topic":"bug/1920099","hashtags":[],"change_id":"I336a8c92bfdd852cab42c9e5f992a30f3632bfaf","subject":"[SVF]: Fix extend issue for a clone of rep-volume","status":"MERGED","created":"2021-03-23 10:48:25.000000000","updated":"2021-08-05 18:16:49.000000000","submitted":"2021-08-05 18:12:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782405-bug/1920099","meta_rev_id":"456c8dfd935448edf831d44fe57593ed6e722a26","_number":782405,"virtual_id_number":782405,"owner":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":0,"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-05 18:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"recommended":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"all":[{"value":1,"date":"2021-07-27 11:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":1,"date":"2021-08-02 12:48:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":0,"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-05 16:44:14.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2021-08-03 12:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":0,"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-05 16:44:14.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":0,"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 10:49:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-23 11:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 11:38:23.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 13:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 13:14:53.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-23 14:34:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 15:50:43.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 16:21:27.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-23 23:39:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 04:53:42.000000000","updated_by":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"CC"},{"updated":"2021-03-24 04:53:42.000000000","updated_by":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"reviewer":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"state":"CC"},{"updated":"2021-03-24 04:53:42.000000000","updated_by":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"CC"},{"updated":"2021-03-24 08:36:12.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 14:45:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-24 19:06:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-26 17:02:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 03:47:44.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 04:21:52.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 04:48:18.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-30 08:26:30.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-30 08:37:21.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-30 12:34:14.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 16:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-03-30 23:02:04.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-31 05:18:20.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-31 06:21:16.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 14:02:33.000000000","updated_by":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"reviewer":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"state":"REVIEWER"},{"updated":"2021-04-05 19:07:00.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-05 21:20:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-04-07 08:30:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-05-03 07:39:50.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2021-05-05 12:41:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-09 05:28:48.000000000","updated_by":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"REVIEWER"},{"updated":"2021-05-09 05:28:48.000000000","updated_by":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2021-06-02 06:13:00.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"CC"},{"updated":"2021-06-08 20:36:55.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 20:07:32.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-07-28 17:45:53.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-07-30 21:12:52.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-08-02 12:48:14.000000000","updated_by":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"reviewer":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"state":"REVIEWER"},{"updated":"2021-08-03 12:44:44.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"7e007cbca6a04bf81e81c0549afa4b79f71b63b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-03-23 10:48:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"322e56f793b0ea4e6e0e84f5622e5d47b4b1bcf1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 10:49:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"fc0ada6b5bec8fe52258f54ad121bcc297a5ddda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-03-23 10:58:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2fd026c79b353338a1372aaade1072459b2d25be","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 10:59:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"30965aca1d314eda11953c20863040a9775a0ce6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:27:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bf34ce586264a3b9219280bd833dd35 : FAILURE in 26m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18df95f8fc574a27a20ac54cb4fccdfd : FAILURE in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dba50346e4227461f087608195dae129e7b63761","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-23 11:38:23.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/782405/782405/2/check-not-vote/ext-cinder-zuul/3287bcf : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ce129cd498143106877c0348ba4c2cf7207bf6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:01:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4175d9129dda43ce8c6be7d7c9bf1623 : FAILURE in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf53072ec5c41db8a8185de2f42e0ec : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51be3a343d18473a84322e4ff6656ef4 : FAILURE in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ba2e5033fc44121839ae5fe3513665b : FAILURE in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf6aed35cd66456b9918cc1870a7947b : FAILURE in 12m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392cadf8157e4d388ddcda4297116104 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/2b6a187447f94d368ea5cd81ee088c3b : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/112dc583c0d64484b0635269fd1b75a8 : SUCCESS in 1h 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/331ee02ee1ba4f04989b483b2c3d3d55 : FAILURE in 11m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e9ba702397c49fba2d4816b349624c1 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e762907510047fab1ca6ba0a9aa7705 : SUCCESS in 6m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/106d06149cc6471297235fff2a1ff2b3 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1871124851c74ba1aa472763a07fcb88 : SUCCESS in 34m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8cbb0ceeed32491eab955c7c83b92e08 : FAILURE in 7m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5485d28d860a4095917dd59d204049da : SUCCESS in 1h 22m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9691a15bf52346999bafa6c61c2881fa : FAILURE in 1h 23m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9cb8b913ad64f94a8323e527e5ba7a4 : FAILURE in 1h 35m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22efee1dac844d2391b8d0aee53007c2 : SUCCESS in 1h 15m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/31af39e619bc43e898811245215c2e0d : FAILURE in 1h 33m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0b19756d5dd64a0bb2ab1c233babe7f1 : SUCCESS in 1h 21m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bef8ec224eae4fbd9c9d9bc40b5e91db : SUCCESS in 1h 59m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2442d6bda6de45ebba99baa4a0c51629 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":2},{"id":"cf00bb58a5e8338ad75be78b282a5908c3b37bee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-23 13:14:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/2/check/dsvm-tempest-cisco-zonemanager-job/fcedc4c/ : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":2},{"id":"aefa470e173c0b4c8d83f9a363f865f30751cfd1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 14:34:49.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/782405/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"5732c5221b6fc91af762dbba5b8b2e77a15a163a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 15:50:43.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/2/check/DellEMC_XtremIO_FC/e943337/DellEMC_XtremIO_FC/1103 : SUCCESS in 2h 03m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/2/check/DellEMC_XtremIO_iSCSI/9622205/DellEMC_XtremIO_iSCSI/1121 : SUCCESS in 1h 43m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"bb4a613dace7435ea60cdcf7a0f6eac17935ae27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-03-23 16:03:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"044b1d167174886c2d261e0343bfd24f704efc8a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 16:05:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"c970fb7a9e24d1bac33c334f122baaece8b2817c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-23 16:21:27.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-1/ns5_nfs : SUCCESS in 5h 18m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-1/ns5_iscsi : SUCCESS in 5h 13m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"30d9fa0e216eb118d0b9f8d49d09e53bb6cc3e00","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-23 16:31:53.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-2/ns5_nfs : SUCCESS in 5h 23m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-2/ns5_iscsi : SUCCESS in 5h 10m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"556ab532f27d0d4aa8fb6e933a3cf31c4d93c88f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 16:32:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0003adb040ae47f1a525cd80052f8803 : FAILURE in 16m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ea3dd8486a34360aeb9564075efe3fe : FAILURE in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ced8469f2c466b453a3039d0dbb858969d07278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 18:27:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d44df8127ea6406c9f6c67756f5bb5d8 : FAILURE in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ad9b7b55464167a6a80fcc5113b1c5 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3703e6ace0d846958c8edd64cd32953d : FAILURE in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8d9252f199d4dc8b6877876c870a4a0 : FAILURE in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aafc6654d0a45ff96befb7d9536772f : FAILURE in 15m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c28d2907904bd7b5b4e8ed7b184600 : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/29362c2f4db640119aa291766b18104f : SUCCESS in 1h 06m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10920af6fda348c4a7ae1367db4a619f : SUCCESS in 1h 20m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aaa60652c90b4e86b020fd66e35d13dc : FAILURE in 13m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e764d80abd31481c867177ff6d45ea8e : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31c4648918e94357a21f63db9d59621e : SUCCESS in 11m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2b4a9d6555949bb9ec61932464f597e : SUCCESS in 7m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b027b1ab121344d994b088d7ae4b5fd3 : SUCCESS in 53m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/775b7ea0783c4b51a7273c6b17da9e9c : FAILURE in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2abca4d54b84469aa065b5d9e0b65608 : SUCCESS in 1h 10m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1138f4c2fb19431896eb8ea8b7bc3b95 : FAILURE in 1h 24m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/334bd5399af94ef8b89ec5deda1aaada : SUCCESS in 1h 32m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9fbfa83850b24d27ad9d2ea278c4bf47 : SUCCESS in 1h 14m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fcc06af835484da0abbbe729076f5247 : SUCCESS in 1h 34m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9bb6e2902437480b8c42c7f2fcea59b8 : SUCCESS in 1h 22m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6805979184c845b1964f6cd99be70b08 : SUCCESS in 2h 06m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d382b3bf4b94dffb33a54c0e523b64f : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":3},{"id":"2b3ea27f5ee2d5923946b83361950d0b11927ee8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 21:19:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/782405/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ddb2bba0ef942af85a246bd14082ecdda9386acd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-23 21:20:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/3/check/dsvm-tempest-cisco-zonemanager-job/055e65d/ : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3e1337c742b4513aa2fa93528412434b588d5b15","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-23 23:39:36.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/3/check/tempest-dsvm-full-focal-py3/278fc4d/ : SUCCESS in 1h 49m 20s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/3/check/devstack-plugin-ceph-dsvm-focal-py3/b520412/ : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":3},{"id":"0b48e5ebd48e490b399a8bcb2a24ddd895581dbb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-24 03:14:02.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/782405/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"7989ffd95a177e36608f6cbddd1fa4157f85da0f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-24 03:37:45.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-3/ns5_nfs : SUCCESS in 4h 54m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-3/ns5_iscsi : SUCCESS in 4h 50m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"f0206a40cb6e5eb503d141de167068fadb3c0d29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-24 06:58:48.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/782405/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a9f363817a21ca731c4dfbe75f1820146ab81e65","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-24 08:36:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/3/check/dsvm-tempest-zadara-driver/858cc84 : SUCCESS in 1h 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ebe5bf1712c0211b255d43d3ec2a50cd253effd","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-24 10:24:49.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/3/check/DellEMC_XtremIO_FC/db93af4/DellEMC_XtremIO_FC/1112 : SUCCESS in 2h 03m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/3/check/DellEMC_XtremIO_iSCSI/80e8e35/DellEMC_XtremIO_iSCSI/1130 : SUCCESS in 1h 42m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"a008cd0243878229eca0e7b20af30a191e7088a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-24 11:50:28.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/782405/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4586bdc5e14b8bd47ad3324af1e43509dcf95cfd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-24 14:45:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/782405/3/check/iscsi-driver-nimble/4e3478a/4e3478a.html : SUCCESS in 2h 12m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c87e86312a0fe9965074c72860b0593a72a73e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-03-24 16:32:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad384a76f29b410ea8a42b6805eaf8fea7a3e4f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-24 16:34:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4f4f62481f76ae26afe384e3bd306e7f9b233f92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:07:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/802810fccfe5494dad344fd318180534 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46b164b869fd42d59d77d663d592245c : SUCCESS in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"577bc5bf0363954ad2adbbd86fd2765bce6bd11b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-24 17:34:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/4/check/dsvm-tempest-cisco-zonemanager-job/d8a57cb/ : FAILURE in 42m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0aa0ec7a0048907a3d8a28d82c7beac508b0949c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-24 19:06:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_Unity_FC/16ed28c/DellEMC_Unity_FC/963 : FAILURE in 5m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_Unity_iSCSI/8605500/DellEMC_Unity_iSCSI/906 : FAILURE in 15m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"5ded236e85d2d2ab2188a6dca1029b801ce3edcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:38:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3897512c6b044085b176beb6a4688b9c : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c623bd3426b04c4ab67ee6becc96a100 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cfb847e3d254ad78e781d3b95f0feb6 : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abbab00567264631b138afe31ea477a9 : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d220033d83fa42d791ac989df9311fb6 : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e9900b07e644179a2bffe36bdfe96c : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/669cba56009847b7931b9b86fcb9054d : SUCCESS in 1h 04m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e1f94a771754d59872f379dc948e636 : SUCCESS in 1h 15m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/099a2874f637485183a1faf0e6922a56 : SUCCESS in 14m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78e7e0604e2d48cc8d2c6aca005b30a6 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a5a83fd45574dce9583620fa19f5ca3 : SUCCESS in 7m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4b4671d55674a008a6b900753afa460 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ec252f8a350414abe3ce3260b756b3f : SUCCESS in 45m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/272d7ca249e14a0ba910ed071e778ee2 : FAILURE in 8m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5b6e3d0dd2ca4e1496ff698e7bd9b945 : SUCCESS in 1h 28m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48708dfaabcf46199c5785d8d4e040b5 : FAILURE in 1h 29m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e75a33723b11413989fa9c3e3d371ad9 : FAILURE in 1h 40m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31d4bcb8d9054a0b8c0e1a183d243a58 : SUCCESS in 1h 27m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/509fde2f63b24e1bb3a5e872e07a7847 : SUCCESS in 1h 25m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e6c5be8549842a7aa88313efdcd4e0e : SUCCESS in 1h 17m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32336225f093430f9ba8c973aff549be : SUCCESS in 2h 07m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afaea05a3e984fe5a5ca40c77e50a19c : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":4},{"id":"d46dce8e27d7945fb69ccaffb645b62483ee1418","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-25 01:50:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/4/check/dsvm-tempest-zadara-driver/466e4f8 : SUCCESS in 1h 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b6e81e3fa4e4bfc2e33b984ff731b9cd9d812b7c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-25 04:56:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/782405/4/check/iscsi-driver-nimble/525f383/525f383.html : SUCCESS in 2h 03m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4ce7b3bd74886c70ef33cd277e38adc152d1b996","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-25 04:59:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_XtremIO_FC/2f8cabb/DellEMC_XtremIO_FC/1121 : FAILURE in 2h 03m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_XtremIO_iSCSI/67cd1e3/DellEMC_XtremIO_iSCSI/1139 : SUCCESS in 1h 48m 59s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"63633018a32e62a2fd3fe3f8b9a4617dc77d4afa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 05:18:52.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-4/ns5_nfs : SUCCESS in 5h 55m 35s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-4/ns5_iscsi : SUCCESS in 5h 26m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"f35b4704314cf4deeb48db0b04d0932a9a018bec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-25 11:03:15.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/4/check/tempest-dsvm-full-focal-py3/24186e6/ : SUCCESS in 2h 08m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/4/check/devstack-plugin-ceph-dsvm-focal-py3/3ef0736/ : SUCCESS in 2h 05m 28s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/4/check/devstack-plugin-ceph-dsvm-focal-py3-newcloud/6d0421e/ : TIMED_OUT in 2h 45m 17s","accounts_in_message":[],"_revision_number":4},{"id":"5ef752aae1d8c98113060b34fcf11141afa6fe52","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 17:02:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/4/20210326233023/NEC-ISCSI : SUCCESS in 2h 17m 00s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/4/20210326233023/NEC-FC : SUCCESS in 2h 32m 01s","accounts_in_message":[],"_revision_number":4},{"id":"7831d849934946eec49df7f72c623a0d7aa31426","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-03-30 01:40:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"47741a0ab41b3fc0702377779cc9f97cda993504","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 02:04:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7eef757497574365aa6bcd46bfd6fc17 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39463758160649a88ac0e9d5f2ad3b61 : SUCCESS in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd2deffe65f21380495146ffe92304f93e315bdd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-30 03:06:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/4/check/dsvm-tempest-cisco-zonemanager-job/022f23e/ : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":4},{"id":"78565cd7f58235582272f89faf95ecccb3b39800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 03:19:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87550199daca45b1919d438da165a6fd : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22af4a3586004b4cb9cf19ad4660dbb5 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4528642e32094075844412ed139e46f6 : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b099033bef4744418452dfc3a590dde5 : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fc0459f53e74403943ad7d878a4719e : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c3d489a3d9435fad36bbc270fea8c0 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/b530f6b62f5341919bcabaed54e6cbc4 : SUCCESS in 56m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eabc4e98806d424ebaf8a0702e06b17d : SUCCESS in 53m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/783e7208c3b5474786307d65d18b8925 : SUCCESS in 12m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12ff4298d9894a7a9588c3852d2daa32 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb9313b8fa2d43cbab0f254fdc37f6e8 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/787bea47b6a8421da905d2b98e4b697b : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7849216e268e48838827f397eb13c202 : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb48fdc06e63498488f5ba37cc3883bd : FAILURE in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa631316f5d2455abb63f5a838acc442 : SUCCESS in 1h 28m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/de1ac9c30aad4cce88347c3f3d113ae5 : FAILURE in 1h 37m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52fcfe20b6cb48989e6a3c1a3947be56 : SUCCESS in 1h 06m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5678c99ef90f49a1b723d74c679a7bfd : SUCCESS in 1h 14m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/806759933d784b88a6a8490c5bbdcd2e : SUCCESS in 1h 02m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f707c26445dd436294b47d958ff72662 : SUCCESS in 1h 11m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/184db0c85b5c4bd4b146914c0ed3a541 : SUCCESS in 1h 31m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/190acc15e219450baac0cf262f0e6826 : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a44e734b8d23675e42d712e471e178e99be7d8f0","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-30 03:46:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/4/check/dsvm-tempest-zadara-driver/a9c5209 : SUCCESS in 1h 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e561c367a027ff1bc4b173b0b8d9ea1c08ac5d53","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 03:47:44.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_PowerMAX_FC/c38e1be/DellEMC_PowerMAX_FC/638 : SUCCESS in 5h 33m 42s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_PowerMAX_iSCSI/79954bb/DellEMC_PowerMAX_iSCSI/607 : FAILURE in 6h 56m 04s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"c2fe8b8bce50d8e69ebb6c0ff135674b37f27b81","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-30 04:21:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/4/upstream-check/cinder-cDOT-iSCSI/2851093/ : SUCCESS in 38m 47s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/4/upstream-check/cinder-cDOT-NFS/dc96091/ : SUCCESS in 43m 39s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/4/upstream-check/cinder-cDOT-FCP/90184b6/ : FAILURE in 1h 00m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1313a5a28f1f4bcde08c4688a53c19b0a3d1919c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-30 04:48:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/4/upstream-solidfire-check/cinder-solidfire-iSCSI/5bfdae8/ : FAILURE in 1h 26m 51s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"0d45370345c78bbf29fd80855ebddc7debf06e7f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-30 08:26:30.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/782405-4/Lenovo-iSCSI : SUCCESS in 49:31.81\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/782405-4/Lenovo-FC : SUCCESS in 43:26.84","accounts_in_message":[],"_revision_number":4},{"id":"c3e1d5d7ccf2afdb84ee4e16e331c74881f113f8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-30 08:37:21.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_SC_FC/fa33144/DellEMC_SC_FC/542 : SUCCESS in 2h 04m 13s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_SC_iSCSI/12ae6a2/DellEMC_SC_iSCSI/539 : SUCCESS in 2h 12m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"454461a4787b89f3170e2c7c04373364fcaaf268","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 11:00:37.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_XtremIO_FC/9b96c1c/DellEMC_XtremIO_FC/1181 : SUCCESS in 2h 04m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_XtremIO_iSCSI/980aa0f/DellEMC_XtremIO_iSCSI/1199 : SUCCESS in 1h 47m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"6c0da7c7a4e0b2534eed66c3fd01173964ab6f36","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-30 11:09:07.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_Unity_FC/db256fd/DellEMC_Unity_FC/1026 : SUCCESS in 1h 40m 38s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_Unity_iSCSI/4dc1133/DellEMC_Unity_iSCSI/967 : SUCCESS in 2h 07m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"32b2a6490ef8d597a99b5cce37253fc604d6c4ec","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 12:34:14.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/782405/4/check/DellEMC_PowerFlex/8e6b736/DellEMC_PowerFlex/692 : SUCCESS in 2h 50m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f2a7134b82a2cb105eadafcbd35913729a36ebfe","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-03-30 16:00:06.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/782405/4/fc : SUCCESS in 0h 53m 41s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/782405/4/iscsi : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":4},{"id":"4e5663846305fa416250018922b08a279942c8b9","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-30 23:02:04.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/782405-4/PVME-iSCSI : SUCCESS in 56:12.93\n* PVME-FC http://os-logs.tristero.net/dell-ci/782405-4/PVME-FC : SUCCESS in 56:56.75","accounts_in_message":[],"_revision_number":4},{"id":"5d53f6f502af997e4e6241863686622b741e0f37","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-31 05:18:20.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/782405-4/Seagate-iSCSI : SUCCESS in 46:11.45\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/782405-4/Seagate-FC : SUCCESS in 39:03.07","accounts_in_message":[],"_revision_number":4},{"id":"52c6417f386827bc01d16efd4159aa5d3acf7557","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 06:21:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/782405/4/check/DellEMC_VNX_iSCSI/64641ff/DellEMC_VNX_iSCSI/674 : SUCCESS in 3h 36m 50s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/782405/4/check/DellEMC_VNX_FC/7fbce6d/DellEMC_VNX_FC/723 : SUCCESS in 1h 29m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"ef8cb6b39592c8aff429b37bb37ed25ad234e246","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-04-01 14:02:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4175b72e2e9d1ed3547352cfac1f0c56a223ec15","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-04-01 16:38:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8eaf918c25bd107fe591890859dfcd6d2222a0f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-04-05 17:57:59.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1bbe1cf043faca81943069be19e775f05b1994d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 18:21:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf923407497f4848beb113b5eb8d7d08 : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee85b14e301543cca35002cdbb4283c1 : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5bf4c0ce909225dcb0c890263b8270ee400d8abe","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-05 19:07:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/5/iscsi/ : SUCCESS in 1h 08m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/5/smb/ : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":5},{"id":"295c28baeafdf41821566356c108b8e7aa9cfd8d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-05 19:22:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/5/check/dsvm-tempest-cisco-zonemanager-job/da46bf1/ : SUCCESS in 1h 17m 45s","accounts_in_message":[],"_revision_number":5},{"id":"c7cedb319b22c72cd48ab9c86c3295ed32225fed","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-05 19:30:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/782405/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/afa10b8/afa10b8.html : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":5},{"id":"0bf1b4924651560e30cbd56284bf36ce10466462","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-05 19:31:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/782405/5/check/iscsi-driver-primera/f1d7575/f1d7575.html : SUCCESS in 1h 32m 21s","accounts_in_message":[],"_revision_number":5},{"id":"7ceac6f3df044f916e54fd3fc1eba9fdd8a91974","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-05 20:12:16.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/5/check/tempest-dsvm-full-focal-py3/92955b8/ : SUCCESS in 1h 46m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/5/check/devstack-plugin-ceph-dsvm-focal-py3/3d14718/ : SUCCESS in 1h 44m 34s","accounts_in_message":[],"_revision_number":5},{"id":"bf4df6efb17796558671312072639812eb49371a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 20:21:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7119dc7875224acf816d38ed8b6fc58e : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2281a5e9774f42f597a14bc0147bf69f : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51d6cdee5ef34abcbecf5a4b2cadd1ee : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08ed3ecd76c54db0baa433b52100fbef : SUCCESS in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9b8e27f58794ca8bdfdfb9bffa360ae : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82fcb514ba3467fbb949b6e10755d7d : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/12af30bb9a034d948c0703b4ee585de0 : SUCCESS in 1h 00m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21c6dbc88ab14e48b6fdaf523513ea6c : SUCCESS in 1h 04m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8f3007c5c5c43389487132449b1535e : SUCCESS in 11m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2d802c76272745ce951e7c0550e0b7ac : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78c0c63ff31c4d11af744cfa0aa65c73 : SUCCESS in 6m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb9dea4eff404604a928410d4aa60f69 : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/042bf8b697104c6f8d150c18cb8f7b40 : SUCCESS in 39m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3aca600434f43ca82b3be0e6de56200 : FAILURE in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3761583ff3d24c958f75ae3197d377ef : SUCCESS in 1h 18m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/954f931db6114918bff5377f68f20bb1 : FAILURE in 1h 25m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8891cba305154609bac004c30af53a98 : SUCCESS in 1h 25m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02d31cc2ce9245dcaa5451a3d7f6d5dd : SUCCESS in 1h 43m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97b707f1610f48fc8687760ce9661aed : SUCCESS in 1h 24m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f3dceb581c1c4ccdadc17420654ebef1 : SUCCESS in 1h 23m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e511bea62d548bdae7cb0a4c97ca951 : SUCCESS in 2h 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4dd45e11b694c1898cc3141f64bbf81 : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":5},{"id":"b58312afc4ad06c59b1a33dbfd99b69c53d2354e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-05 20:47:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/5/check/dsvm-tempest-zadara-driver/b73db7c : SUCCESS in 1h 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c60071306b5e3482b54499c7b3db53ee72969a1a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-05 21:19:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/5/upstream-check/cinder-cDOT-iSCSI/5c2aeca/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/5/upstream-check/cinder-cDOT-NFS/a6b02fe/ : SUCCESS in 39m 12s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/5/upstream-check/cinder-cDOT-FCP/a0d6c75/ : SUCCESS in 57m 07s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"5767513e407c11139a7b70ed9ad261de1294fc5a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-05 21:20:03.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/05/782405/5/check/DellEMC_PowerStore_iSCSI/fd67e92/DellEMC_PowerStore_iSCSI/739 : SUCCESS in 58m 24s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/05/782405/5/check/DellEMC_PowerStore_FC/a957829/DellEMC_PowerStore_FC/734 : SUCCESS in 57m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"6d5107eea5a0a2522d01f690337ba2004847b29e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-05 21:36:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/782405/5/check/DellEMC_VNX_iSCSI/7ce3f8d/DellEMC_VNX_iSCSI/709 : FAILURE in 1h 15m 23s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/782405/5/check/DellEMC_VNX_FC/d1e6772/DellEMC_VNX_FC/758 : FAILURE in 1h 01m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"6504714bb6b8d45604b8cfa8ced4ec1f4076ac5d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-05 22:05:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/5/upstream-solidfire-check/cinder-solidfire-iSCSI/7f1f882/ : FAILURE in 1h 43m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"036908db762df16a35b31ce2c76b185bcc4183ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-05 22:20:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/782405/5/check/iscsi-driver-nimble/5e38b72/5e38b72.html : FAILURE in 1h 22m 48s","accounts_in_message":[],"_revision_number":5},{"id":"e9ee40f1862739a7c7c74d6e6195d72e1ab82409","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-04-05 23:00:04.000000000","message":"Patch Set 5:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/782405/5/fc : SUCCESS in 0h 45m 27s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/782405/5/iscsi : FAILURE in 1h 07m 53s","accounts_in_message":[],"_revision_number":5},{"id":"ab5e961f5e025b89bb0346c963d51ee25bf8cc6e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-05 23:06:00.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-5/ns5_nfs : SUCCESS in 5h 7m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-5/ns5_iscsi : SUCCESS in 5h 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1e8294cd2cb9a5ae9ba277b3b914d818c47bfb6b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-06 00:15:03.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/782405/5/check/DellEMC_SC_FC/75879b3/DellEMC_SC_FC/581 : SUCCESS in 2h 01m 46s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/782405/5/check/DellEMC_SC_iSCSI/cf183fb/DellEMC_SC_iSCSI/578 : FAILURE in 2h 20m 53s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"36bbc31641598e65f1137d215d4ab8b1cb7316e1","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-06 00:24:40.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/782405-5/PVME-iSCSI : SUCCESS in 53:26.61\n* PVME-FC http://os-logs.tristero.net/dell-ci/782405-5/PVME-FC : SUCCESS in 56:24.07","accounts_in_message":[],"_revision_number":5},{"id":"75006cac337bea03fad95c4c26ccbd313b541536","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-06 00:29:07.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/782405/5/check/DellEMC_Unity_FC/3b9f24c/DellEMC_Unity_FC/1085 : FAILURE in 2h 00m 00s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/782405/5/check/DellEMC_Unity_iSCSI/8615104/DellEMC_Unity_iSCSI/1022 : FAILURE in 1h 29m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"5526cb79136e435d56e1982aa784991a7837f253","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-06 01:19:09.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/5/check/DellEMC_XtremIO_FC/1505cac/DellEMC_XtremIO_FC/1236 : SUCCESS in 2h 06m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/5/check/DellEMC_XtremIO_iSCSI/a66c482/DellEMC_XtremIO_iSCSI/1256 : SUCCESS in 1h 48m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"a489bd9b3fac2d74327447b2e37ebab4765232ce","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-06 01:51:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/782405/5/check/DellEMC_PowerFlex/cc4376b/DellEMC_PowerFlex/730 : FAILURE in 2h 50m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"d4a23dabb7d743bd8f3dede1ae44cd9e6487133a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-06 02:37:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/5/20210406071921/NEC-ISCSI : SUCCESS in 2h 05m 39s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/5/20210406071921/NEC-FC : SUCCESS in 4h 17m 55s","accounts_in_message":[],"_revision_number":5},{"id":"927e3603474a866c37f539316d8010b774ffa82a","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-06 18:47:20.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/782405-5/Lenovo-iSCSI : SUCCESS in 48:01.32\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/782405-5/Lenovo-FC : SUCCESS in 42:18.75","accounts_in_message":[],"_revision_number":5},{"id":"83b59d074d1144610081e8e26e88b7796d37c2c8","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-07 08:30:31.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/782405-5/HPMSA-iSCSI : FAILURE in 1:02:14\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/782405-5/HPMSA-FC : SUCCESS in 34:38.18","accounts_in_message":[],"_revision_number":5},{"id":"b05d2c58a62b69f66fc1213d0865efb88126b2fd","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2021-05-03 07:39:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c6cff1d41cd7c08735dd24e58411368090f04e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-05 12:09:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c7ede264247dc963a8cee1422127180c6706058a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-05 12:41:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/6/check/cinder-storpool-tempest/163ac2f/ : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":6},{"id":"84eba560ff9e08e5588b74088064b03ec99b0531","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:06:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77ff2c8cfe5249beaed14348c8c685be : SUCCESS in 22m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/931a6437ba1948068b1b8aae72187e06 : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"67c253cf7cdef7b6393e4efa15ff3115c50457ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-05 13:10:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"53df2eab5175b5607383628aec7a9002d621d102","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-05 13:23:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f2138f70cbc08894db31a93d6409fe6a1925cac9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-05 13:42:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/7/check/cinder-storpool-tempest/4e4ced4/ : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":7},{"id":"f3d691ce46e5fd6a40b232bd08c0ea8740f5e826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:44:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e929dda42334e3dae44bc9a439440ee : FAILURE in 27m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8556cce67cc40f394fdcfdad0535214 : SUCCESS in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3809a2fccc0ba2253928cbd279e2d7d046102374","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-05 14:01:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/7/iscsi/ : SUCCESS in 50m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/7/smb/ : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":7},{"id":"78b11feb92c8e462c0e00efb0abb1d7a95cb9fb7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-05 14:22:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/782405/7/check/iscsi-driver-primera/c89d721/c89d721.html : FAILURE in 1h 11m 19s","accounts_in_message":[],"_revision_number":7},{"id":"75636ab235278bf848f889e7f673916415fd1a04","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-05 14:28:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/7/check/dsvm-tempest-cisco-zonemanager-job/9dde580/ : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":7},{"id":"fe34a8691dd13934aa67b61d5fd2cd8d52ec8f26","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-05 14:32:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/7/check/dsvm-tempest-zadara-driver/67ec0f8 : SUCCESS in 1h 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e6f82551b9a3c2e60f02c6d41da65fcbbfe43865","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 14:39:22.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/6/20210505210956/NEC-FC : FAILURE in 48m 15s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/6/20210505210956/NEC-ISCSI : SUCCESS in 2h 29m 21s","accounts_in_message":[],"_revision_number":6},{"id":"9dbb731c00fd2c5c5fc04e31e64e71477eaa7b55","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-05 14:51:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/339722d/ : SUCCESS in 1h 39m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4fa2a6d/ : SUCCESS in 1h 32m 42s","accounts_in_message":[],"_revision_number":7},{"id":"2911efa15b076f0b672c63e753b65962ebda8065","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-05 15:09:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/782405/7/check/iscsi-driver-nimble/b44367d/b44367d.html : SUCCESS in 1h 54m 50s","accounts_in_message":[],"_revision_number":7},{"id":"ee1c03744add936b8f95c8d0f7393158af688ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:15:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2e996162eff44229ea1b7579d087ded : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3e0c3f27a64693b650e7e876d061d4 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4530d4648604d3e985c09dbe92800e5 : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/203697e5991d451c96c811fe2db331e2 : SUCCESS in 10m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef7659d185d341dbb7f7feb99c4044ea : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8164efe4664b4bd6b51715436c702819 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/1d886f82f610470eb894050fbad522ad : SUCCESS in 1h 08m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8e6a38599f04b0eaa9d85fc43d54053 : SUCCESS in 1h 17m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8c314838d884658966c178cc3689ca9 : SUCCESS in 15m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2d14701e17844b4abf64a4da7346168d : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29ea33657ce24b6ca5007cc5a4e7df93 : SUCCESS in 7m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a5ae9bf7b19432e81733b763d14362e : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/81d63bc73c474c2f9dec1273ec352618 : SUCCESS in 40m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7529f99981845309127703b9e58fec9 : FAILURE in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8490dd57f144b61a4528e1f7d5b5fa1 : SUCCESS in 1h 08m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/67b71db81b554e7986066e325fc65e22 : FAILURE in 1h 25m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bc4a88c9abb142358344c0b5cf1fcc8e : SUCCESS in 1h 23m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9435a1b246904264a2e9567f7e373227 : SUCCESS in 1h 11m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dfb8bdf1f306461292b9563703dfce4f : SUCCESS in 1h 07m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c864f85a31b40758be173dca98790ee : SUCCESS in 1h 08m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d25d6b4b9c9242dc899fbd180e0fa897 : SUCCESS in 2h 03m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ce4922a73794663834a3348bdce18df : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":7},{"id":"7fb1bdf970c10d4a06155a4c92ff0bc4f583fb18","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-05 15:56:22.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/782405/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/df124d5/df124d5.html : FAILURE in 2h 44m 41s","accounts_in_message":[],"_revision_number":7},{"id":"f87d309fb13a27825cfc53c1c712c83b46a7cd43","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-05 16:01:36.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerStore_iSCSI/d6fd199/DellEMC_PowerStore_iSCSI/48 : SUCCESS in 45m 45s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerStore_FC/ed60e15/DellEMC_PowerStore_FC/49 : SUCCESS in 46m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"b478c6104c21ce0ced20f7e6dbfa64249ef4fa55","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-05 16:14:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-iSCSI/7dc747d/ : SUCCESS in 40m 09s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-NFS/b569e52/ : SUCCESS in 40m 28s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-FCP/68c4fbd/ : SUCCESS in 58m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"2bc63d491a8a17a1f0f7209d9f66f83699808ab0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-05 16:16:06.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/7/upstream-solidfire-check/cinder-solidfire-iSCSI/89231b6/ : FAILURE in 58m 20s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"97719d05d559ff7a925bf8cd58d1572968c706bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-05 16:35:33.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_Unity_FC/2b8d375/DellEMC_Unity_FC/119 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_Unity_iSCSI/ac487ac/DellEMC_Unity_iSCSI/110 : SUCCESS in 1h 31m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"93330ae09ddb09f222708a510a45a65ae344cc83","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-05 16:39:10.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-6/ns5_nfs : SUCCESS in 4h 28m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-6/ns5_iscsi : FAILURE in 53m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"8b44009c6b9fe47cac53d470a375eca7df15dbfe","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 17:05:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/7/20210505233933/NEC-FC : SUCCESS in 2h 18m 30s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/7/20210505233933/NEC-ISCSI : SUCCESS in 2h 26m 21s","accounts_in_message":[],"_revision_number":7},{"id":"539008767e4f63175c33dbfd192b7759b68c4069","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-05 17:23:43.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/782405-7/Seagate-iSCSI : SUCCESS in 43:05.28\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/782405-7/Seagate-FC : SUCCESS in 37:15.25","accounts_in_message":[],"_revision_number":7},{"id":"00c6564175d8fde3f5c1bda4d47a76f921b82c1e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-05 17:39:06.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-7/ns5_nfs : SUCCESS in 4h 22m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-7/ns5_iscsi : SUCCESS in 4h 28m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"b163e8b4ab81c4f54846b6575c7a1f45dd73ce04","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-05 17:50:01.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/782405-7/Lenovo-iSCSI : SUCCESS in 46:15.66\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/782405-7/Lenovo-FC : SUCCESS in 42:45.23","accounts_in_message":[],"_revision_number":7},{"id":"baa1eb59e375b0217ede516c89514cb965b01938","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-05 17:50:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"69f5a455d884f535e1dcdae787c7f93c35ffe9b7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-05 18:07:40.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/7/check/tempest-dsvm-full-focal-py3/bc22f78/ : SUCCESS in 1h 52m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/cb854658b79c4a84ba65f045d67c3d12 : RETRY_LIMIT in 27m 48s","accounts_in_message":[],"_revision_number":7},{"id":"fde54fac3987badcd716019691b44ee73a3ab2b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 18:14:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efe20d2cb2cd4dfeabef3082670aac01 : SUCCESS in 21m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2f15ec91cb949569754d8f752c4df25 : SUCCESS in 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4ca3f404cc5d34e6a6bdc67f5e84d60eb74de965","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-05 18:28:16.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_SC_FC/de74735/DellEMC_SC_FC/71 : FAILURE in 1h 58m 05s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_SC_iSCSI/fcab141/DellEMC_SC_iSCSI/68 : FAILURE in 2h 13m 59s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"e50d85cf122cd47dfcc9cdcea86cf8fd5ec6b880","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 18:28:43.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_XtremIO_FC/4f701bd/DellEMC_XtremIO_FC/126 : FAILURE in 2h 12m 35s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_XtremIO_iSCSI/d3be6e7/DellEMC_XtremIO_iSCSI/126 : SUCCESS in 2h 01m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"50481995ad9bb867ff3916a02fe26ec8ef5ed9ec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-05 18:29:22.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/7/check/cinder-storpool-tempest/b8d729e/ : POST_FAILURE in 37m 05s","accounts_in_message":[],"_revision_number":7},{"id":"2800e82e645d7cc289ad2655114989bb29c637ad","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-05-05 18:39:56.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/782405-7/PVME-iSCSI : SUCCESS in 53:22.53\n* PVME-FC http://os-logs.tristero.net/dell-ci/782405-7/PVME-FC : SUCCESS in 53:59.60","accounts_in_message":[],"_revision_number":7},{"id":"94e6d771b40b92453d7af3f7c939050b225110fc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-05 19:03:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/7/iscsi/ : SUCCESS in 1h 11m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/7/smb/ : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":7},{"id":"33b4747d3aa1e83529110da393ebf657ce03ca10","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-05 19:04:20.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerStore_iSCSI/e5353b4/DellEMC_PowerStore_iSCSI/49 : SUCCESS in 47m 24s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerStore_FC/fad05ab/DellEMC_PowerStore_FC/50 : SUCCESS in 49m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"5a37a6b86e558d7fe5f58b68726dec80e49acc51","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-05 19:08:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/7/check/dsvm-tempest-cisco-zonemanager-job/cfd5169/ : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":7},{"id":"b9c4da8a900362d0c4ab426a45ff6e6af09c39f7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-05 19:10:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/7/upstream-solidfire-check/cinder-solidfire-iSCSI/d3ca266/ : FAILURE in 54m 58s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"72f91eb4d301abafeef5f8a6b986762e352b5173","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-05 19:19:38.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/782405-7/HPMSA-iSCSI : FAILURE in 1:26:24\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/782405-7/HPMSA-FC : SUCCESS in 32:40.82","accounts_in_message":[],"_revision_number":7},{"id":"bd19d116ffed3167a18d66e9506eea357f85ca99","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-05 19:21:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2933597/ : SUCCESS in 1h 28m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d0a3251/ : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":7},{"id":"d251e7b5008f9477cd95bfa17b3d9e49fddf5ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:31:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af110d8721d343798339729f8b7c2975 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ef4cce7a344f248c63cf6a830d7758 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dd8be7ac57e4814be79b769d484d376 : SUCCESS in 10m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0798f3b36924890b6ab08694da63b7e : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c2e2844b538441e92b26830b9d8cf87 : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f57567fe1f74479b721e18c10523e87 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/68e04b848ede4d9fb6f7523fde6a47ba : SUCCESS in 56m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a4b44d18bc346059fca14f764ac33c8 : SUCCESS in 1h 06m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21d29e2e106343378a6301823fcb8acb : SUCCESS in 14m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a71144b1e0864551b3999f5ffbef09a3 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6b8427388624731bb63cab00c2bac52 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e71e525da609437d9df6791f19e59d3a : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f712d3dafaba43ccabd6c820b95ee000 : SUCCESS in 35m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e77f74243094c61be59d8c006ef35fc : FAILURE in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c02d8b59ae146fea51b43cc017b7fa5 : SUCCESS in 1h 37m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b3904a8725a4e49af0595fa62cf3376 : FAILURE in 1h 19m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc6aa9999bc54fd8982905a8694d265d : FAILURE in 1h 09m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f975e026668491aa937023fc77937e6 : SUCCESS in 1h 12m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e05239b111a49b2969a261f377a301a : SUCCESS in 1h 27m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e33cc3bc7bfb4135b94214a2020a4000 : SUCCESS in 59m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b3c50b90d0c4741915c4e30df11d0b7 : SUCCESS in 1h 36m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef1929100fd04e5a87e4a507c489de38 : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":7},{"id":"6adc4cf07778b04b3cdc65400a28cc85b078b0b4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-05 19:49:30.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_Unity_FC/8a70703/DellEMC_Unity_FC/120 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_Unity_iSCSI/cdc182a/DellEMC_Unity_iSCSI/111 : SUCCESS in 1h 32m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"0016e1aed4dfe6ebbebc6a7bab4711eda6521806","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 20:06:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerFlex/a515f75/DellEMC_PowerFlex/51 : FAILURE in 2h 44m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"ce4e07b60db419fa56be386400535da55211a1fa","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-05 20:31:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/7/check/dsvm-tempest-zadara-driver/28e91ec : SUCCESS in 1h 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0b09ed18e0593a55c09a3b930f0c01ffdef62930","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-05 21:03:17.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-iSCSI/03567f5/ : FAILURE in 35m 26s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-NFS/e96fd4d/ : SUCCESS in 36m 41s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-FCP/03e9754/ : FAILURE in 2h 47m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"03b5cb6c2f71112228902bd0827eb543a0b2ddcb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 21:18:49.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_VNX_iSCSI/2782e30/DellEMC_VNX_iSCSI/75 : SUCCESS in 2h 59m 14s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_VNX_FC/b0d94e0/DellEMC_VNX_FC/78 : SUCCESS in 1h 20m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"9b149dc2e9d05a80e778dc4b7e62b64ee9329435","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-06 07:39:01.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"1c60f1401cd849ff27d14994a0ff81e283584820","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 08:08:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e947f1a46d924d91b15693ea7b08a6d2 : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7083ce152d3422a9ef576b1e872b079 : SUCCESS in 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"30bf7f02d2beaeeeb82fe3849064989a797f3909","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-06 08:13:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/7/check/cinder-storpool-tempest/3a7534e/ : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":7},{"id":"c1d88fdae5121fb0f389327c50bcb474696375c0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-06 08:30:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/7/iscsi/ : SUCCESS in 50m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/7/smb/ : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":7},{"id":"dcdb7a1b1f2d13cad307e38658ee8770fe7c51fd","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-06 08:51:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/7/check/dsvm-tempest-zadara-driver/c843b61 : SUCCESS in 1h 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"443ea87bd553f26a4fa9a3a8eb082ac6fa13bcd5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-06 08:54:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/7/check/dsvm-tempest-cisco-zonemanager-job/dda7549/ : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":7},{"id":"0a6f9c2f9a00953b03f2586e43a49711bb6ffcbd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-06 09:25:17.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/26042ed/ : SUCCESS in 1h 42m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ad50030/ : SUCCESS in 1h 30m 40s","accounts_in_message":[],"_revision_number":7},{"id":"4a094797226d46ece608bc5e3f7a208f9e79f3e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 09:40:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1099a760432943ebb375543dc91c3d1b : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1457d710cd247b692f3c68d9774f660 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ba1ef1cc10428e85a08fff76d6bae1 : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/683be297a8e24ddba76990714329e900 : FAILURE in 14m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb04dae71f28462690c05fe6f79c051a : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbd0001306449c2b9e752d96bcfb146 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/ac2c50b689224869bd353502e3c083a2 : SUCCESS in 1h 00m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6fce972966fa4b3d8497cb1b36b78f66 : SUCCESS in 57m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/19db93df04c54b0e9835279c149d3dbc : SUCCESS in 13m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a01d09262cd24a6ea91fb168d768cd09 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3f7c3c9492e46199e8a9591e0a7d3d3 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e6d6ea30cb248408046c31007f883a0 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28a8b10cbc884fef940bdc267253f94a : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab47232efdb34ea4816bf151efe864c2 : FAILURE in 8m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59a9f5d401224e93949830dc0e411a57 : SUCCESS in 1h 09m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e9555dda0214bf387571d5a053222b2 : FAILURE in 1h 34m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d9573aa43b14f6985a3df4e13ac4161 : SUCCESS in 1h 38m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fb1e86b0063246b1ab2f5ca175a768d6 : SUCCESS in 1h 15m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3c0d0b166d2e4ac680c7c5772a61fa28 : SUCCESS in 1h 23m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/61c057c8189147d2acbe57090dffe359 : SUCCESS in 1h 13m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cfc4abd975d43d081edb01485bf30a5 : SUCCESS in 1h 57m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26a306a8f11d472cb37e7c0c782e6dbe : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":7},{"id":"a1f3d4c1d533f8102cc94e447b4f598c88be998f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-06 11:16:19.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/7/check/tempest-dsvm-full-focal-py3/139fd4e/ : SUCCESS in 1h 34m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/241c47edb3cd4775b75d3d40fb207044 : TIMED_OUT in 3h 11m 08s","accounts_in_message":[],"_revision_number":7},{"id":"c37e0a2d6bc55001958b0f55503237541ccdbd53","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-08 14:15:00.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"577fd0c0476406508c0442b7c134f944fb7ac086","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-08 14:26:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/7/check/dsvm-tempest-zadara-driver/864fef0 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f3938b50f59c9e45094543f64d94982e9c7a66cc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-08 14:29:30.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/7/iscsi/ : FAILURE in 13m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/7/smb/ : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":7},{"id":"ec18cae7faed23dcbea06e43b14fd73f25b9274f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 14:35:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc88460312854a22a445313ea10bdc29 : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfd344cc409b4485b3344016d1631d21 : SUCCESS in 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aca9270e4260108a888018d36814282af8c70257","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-08 15:33:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/7/check/dsvm-tempest-cisco-zonemanager-job/131ce67/ : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":7},{"id":"2e5a99e726a36a5311de31ba2891223ba3f383b2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-08 15:43:09.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7dd6f10daf9346e2a205ee5d7db46712 : POST_FAILURE in 1h 19m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/6dc4320938e541a7b13b216758e0574f : POST_FAILURE in 1h 24m 10s","accounts_in_message":[],"_revision_number":7},{"id":"896d8920b2cb5b922482b6fa6f142693cb1ed2aa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-08 15:56:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/122e774/ : POST_FAILURE in 1h 40m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/40f7fbd/ : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":7},{"id":"b64b8c3762714bec9507d796dc83f0cbd62601dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 16:43:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17b8880b809c42288489df76bd10425c : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486a03a0889b4d8eb39d49c194c7df63 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb56c2654f94b1b82f98bfedea6cb31 : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7a2e3f43db84c62b34bcf8ea591379b : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cd7e5766b5f44a496458ae616aade3e : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c90d1246f6354ce38134ce42c3291b8f : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/89389082bca94494b49f0839f050f8bd : SUCCESS in 56m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df329af75ee3417b96fede8c19e919b5 : SUCCESS in 1h 36m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cd82b84e530b4c04950b699227641de5 : SUCCESS in 12m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9aa7cd09336b4a3594633e37c230d9e7 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b90b1e02cb24eb1ad04974b5e081224 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79672a5c8d9d466eb2f72545e24536c2 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed3d2b8584404883a8dc60ad6c18e73f : SUCCESS in 33m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee8ebebf0b3b47da9d7ec8c31610ed00 : FAILURE in 8m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec29a14c32ad47fdbd177f306fc61337 : SUCCESS in 1h 16m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0c346759bb394a398729345a34bc051c : SUCCESS in 1h 21m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ab5249c9ac0f4440af788003990ea042 : SUCCESS in 2h 26m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c795645d31f142a581d617b5f0d7b8f0 : SUCCESS in 1h 01m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/414e03d7c0e541e1ab45cddc92f5475f : FAILURE in 1h 08m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ecd474928ab0471c9f16d7989007b4a0 : SUCCESS in 1h 14m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d86349a0c3f842648e2896dcc37e7cd7 : SUCCESS in 1h 53m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b54bd15c96e5405c95f369deee706ef1 : SUCCESS in 47m 11s","accounts_in_message":[],"_revision_number":7},{"id":"45ff13fc9f97853106cc8dec31dbd9dd81f166c7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-08 17:29:50.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerStore_iSCSI/9cf2345/DellEMC_PowerStore_iSCSI/57 : FAILURE in 43m 19s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerStore_FC/49e5487/DellEMC_PowerStore_FC/58 : SUCCESS in 45m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"829f4d0f18c21610f954133361774d4af2c10595","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-08 18:16:58.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerFlex/8ca8d9c/DellEMC_PowerFlex/65 : SUCCESS in 1h 33m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"d47609cff347539921e340541d8b6f7088248419","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-08 18:19:07.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-iSCSI/e63ce48/ : SUCCESS in 1h 34m 41s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-NFS/563d5c2/ : SUCCESS in 35m 24s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/7/upstream-check/cinder-cDOT-FCP/6871139/ : SUCCESS in 57m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"120c2cf42341718131e70e277813ec1633b41976","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-08 18:27:10.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/7/upstream-solidfire-check/cinder-solidfire-iSCSI/81a12cb/ : FAILURE in 1h 42m 22s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"f08c9992df3faf25427bb01fc88185e264f56b75","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-08 18:27:33.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_Unity_FC/fe86a93/DellEMC_Unity_FC/132 : SUCCESS in 1h 20m 10s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_Unity_iSCSI/7fb439e/DellEMC_Unity_iSCSI/122 : SUCCESS in 1h 43m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"7b2db0c4e0ac601d1533c418b2e69c9868fc7503","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-08 18:35:45.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/782405-7/Seagate-iSCSI : SUCCESS in 44:31.38\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/782405-7/Seagate-FC : SUCCESS in 38:15.04","accounts_in_message":[],"_revision_number":7},{"id":"7060060a80d0aa9d3563c13f9d5ad40b21d9bef5","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-08 18:41:23.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/782405-7/Lenovo-iSCSI : SUCCESS in 45:23.92\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/782405-7/Lenovo-FC : SUCCESS in 43:08.82","accounts_in_message":[],"_revision_number":7},{"id":"95fc0a12fb559b72fb09f10303bfe35cc08ea8d1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-08 18:45:31.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_SC_FC/5409d7e/DellEMC_SC_FC/79 : SUCCESS in 1h 59m 58s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_SC_iSCSI/8d584d7/DellEMC_SC_iSCSI/76 : SUCCESS in 2h 01m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"8fb9b3ee0b916c2099706b44b1fb8da014fc97cc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-08 18:52:33.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_XtremIO_FC/174f451/DellEMC_XtremIO_FC/139 : FAILURE in 2h 08m 44s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_XtremIO_iSCSI/f349d5e/DellEMC_XtremIO_iSCSI/139 : SUCCESS in 1h 56m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"8f70951fa548691e50645da109db87c991fe5cd0","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-05-08 19:00:20.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/782405-7/PVME-iSCSI : SUCCESS in 54:47.69\n* PVME-FC http://os-logs.tristero.net/dell-ci/782405-7/PVME-FC : SUCCESS in 54:02.19","accounts_in_message":[],"_revision_number":7},{"id":"49985b11e49af172eb16fdf1894d12bdb1928f69","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-08 19:12:42.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/782405-7/HPMSA-iSCSI : FAILURE in 1:26:47\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/782405-7/HPMSA-FC : SUCCESS in 33:16.69","accounts_in_message":[],"_revision_number":7},{"id":"223c7f22bbccd734b5554b340a73d68e9af69501","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-08 19:55:58.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_VNX_iSCSI/e9b432d/DellEMC_VNX_iSCSI/83 : SUCCESS in 3h 12m 12s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_VNX_FC/912ed25/DellEMC_VNX_FC/86 : SUCCESS in 1h 23m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"08d030f17498a039efa0b4215385daf443a279cb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-09 00:23:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/7/check/cinder-storpool-tempest/c26520c/ : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":7},{"id":"87a747c84087d127087c925ea922325774c56bf9","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-09 05:28:18.000000000","message":"Assignee added: Amar \u003camar.rajeev.singuluri@ibm.com\u003e","accounts_in_message":[],"_revision_number":7},{"id":"728ff5422551b38fde5f67bed287f90aa0317698","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-09 06:32:46.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerMAX_FC/ab6cd5a/DellEMC_PowerMAX_FC/71 : FAILURE in 5h 28m 01s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/05/782405/7/check/DellEMC_PowerMAX_iSCSI/2d786ff/DellEMC_PowerMAX_iSCSI/60 : FAILURE in 9h 53m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"6373e2c44210776c91cfb3a59d723aa6a331f3fc","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-13 09:50:55.000000000","message":"Assignee deleted: Amar \u003camar.rajeev.singuluri@ibm.com\u003e","accounts_in_message":[],"_revision_number":7},{"id":"0e744c80f7e254cb6809c0e494c18784c7aadde4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-19 05:06:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9ad2334ebb681ec84fd68f984307d4782f4782ce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-19 05:14:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/8/check/cinder-storpool-tempest/95b3d6a/ : FAILURE in 7m 02s","accounts_in_message":[],"_revision_number":8},{"id":"c7afacf6a592b20385605e3610c14cf43addf42f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-19 05:22:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/8/check/dsvm-tempest-zadara-driver/2359c03 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c67fa1d0e115b447778591388414d88ffa142e19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 05:29:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8443972ed1634803a75b3a2866518420 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aca61a4f5c7a467298c368ed3bd8bedd : SUCCESS in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e9308d85e2adf3a5b2d25c238e5cb7b8559be947","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-19 05:31:14.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-8/ns5_nfs : FAILURE in 21m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-8/ns5_iscsi : FAILURE in 24m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"4fe23ebb0e7705759e362fc567696a897e969dff","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 05:33:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/782405/8/check/iscsi-driver-nimble/99aefe4/99aefe4.html : FAILURE in 25m 48s","accounts_in_message":[],"_revision_number":8},{"id":"a46d7e7c22eb3786c94d90135dce23f72304d186","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-19 06:08:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/8/iscsi/ : SUCCESS in 1h 00m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/8/smb/ : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":8},{"id":"98a09ed4bab756e45056f3ed1bf47ac7d52177ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 06:18:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/782405/8/check/iscsi-driver-primera/02b78a0/02b78a0.html : FAILURE in 1h 11m 19s","accounts_in_message":[],"_revision_number":8},{"id":"9e6c01705e8a0e9660a3bf262928b74e2760fa05","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-19 06:23:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/8/check/dsvm-tempest-cisco-zonemanager-job/207f8cb/ : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":8},{"id":"d84a4dacfc7b3fae7abab84e5571eb8993fe40f5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-19 06:45:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b05a005/ : SUCCESS in 1h 37m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/469253a/ : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":8},{"id":"603d1a162334d4d4a9bb376efdbd7573c1c9f6aa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-19 06:51:03.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7666833186c04d358a5dd6495e26b830 : RETRY_LIMIT in 35m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/a5da186be9434434a42059e544db4361 : RETRY_LIMIT in 36m 13s","accounts_in_message":[],"_revision_number":8},{"id":"eaa9a61e6cae53de63b7e92c8fcda21aca5c694e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 07:23:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adcfc58ac3744474a6ca68cd459be2be : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4c1805aba7430caca24222a6dfeb55 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d94a5d12bbe6488ba174b437a04a28f2 : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/035322964e1d4fc2be6f373e5735dd1e : SUCCESS in 11m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c976fdc80a3a4a798c67a78f103e559d : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e608d928946488696c413ae083971e0 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/780379dc2b3c434cbdeee422fda52540 : SUCCESS in 54m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0173f581c09144fb82f8765889e3e34a : SUCCESS in 1h 08m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a776f40d73cf44e8bfb052fdd4b3d4a7 : FAILURE in 6m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21d9dc65afb0463982af92384d585fa8 : SUCCESS in 12m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b65d3831ad2b488e859123524ad5d17a : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/170926e4d42941e4a64c96b5288d1653 : SUCCESS in 6m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed1ea98c92a847d98a292715bbb02d5f : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4f34369f939140d2af0b8a8376664e65 : SUCCESS in 38m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec2a4537e83c405ab3c83d81695322a3 : FAILURE in 7m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1408dcd3dd3248878e27f8f90df4048d : SUCCESS in 1h 19m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/99edf2b9bcd240a0978a4c6d00407dd7 : SUCCESS in 1h 22m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ad38c2a57f04ab38379b67686316209 : SUCCESS in 1h 45m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/21ee6f4df18d4b16ab1ff91aa86c4b9b : SUCCESS in 1h 14m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9b8132c6490b4aa68f95f7eab8652ae1 : SUCCESS in 1h 31m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b5468e1ab7144359f534d89fb0c0aa6 : SUCCESS in 1h 21m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32b55a1c7f754fefac75e117602a8cc9 : SUCCESS in 2h 09m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68c34a967341476cbb2fe0f7413bf6fe : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":8},{"id":"a5a508889d19b4bb2d2d4d8f585b00ccbf556334","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 07:31:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/782405/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f3db52e/f3db52e.html : SUCCESS in 2h 23m 36s","accounts_in_message":[],"_revision_number":8},{"id":"23c79e7679830aa9ee26f153790a394d54e25382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-19 16:15:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"73498212935b329bda7eae2b6dcc46a3ab10bde8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-19 16:25:03.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/9/check/cinder-storpool-tempest/c42fd21/ : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":9},{"id":"f7a9cf5822531c5f1c7abc971a45e2a4876e9ef4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-19 16:31:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/782405/9/check/dsvm-tempest-zadara-driver/b50e8de : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"99a5faf7694aa5383b19037819b920f46eff3af4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:41:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99e0bf6eada94050879915ff66aab47f : SUCCESS in 21m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23d79c0a30d546a9ae3e1f2fd6a15d86 : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"12976b9ff50839da8cf9ec3c0d2deaa3b972c8c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 16:43:02.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/782405/9/check/iscsi-driver-nimble/1965d88/1965d88.html : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":9},{"id":"6da6e32e2b1bdc99dd36c381c391969fef75b9c8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-19 17:22:33.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/9/iscsi/ : SUCCESS in 54m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/9/smb/ : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":9},{"id":"7197c0e5478035ae070652c81471a27126ae2673","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 17:27:54.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/782405/9/check/iscsi-driver-primera/96faf94/96faf94.html : FAILURE in 1h 10m 50s","accounts_in_message":[],"_revision_number":9},{"id":"7145e8d16539687fca205a856a74557689a623a7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-19 17:35:52.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/9/check/dsvm-tempest-cisco-zonemanager-job/a8193f2/ : FAILURE in 1h 15m 14s","accounts_in_message":[],"_revision_number":9},{"id":"40d9a45c10197accbddcc6640b031272fc8a2d4f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-19 18:00:30.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/9/check/tempest-dsvm-full-focal-py3/009d49f/ : SUCCESS in 1h 27m 28s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/9/check/devstack-plugin-ceph-dsvm-focal-py3/5cd3b3d/ : FAILURE in 40m 29s","accounts_in_message":[],"_revision_number":9},{"id":"e606d063771a403f0e7bc819c801d923950e9e9f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-19 18:13:08.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d7d32be/ : FAILURE in 1h 56m 25s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/075b031/ : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":9},{"id":"e7905843f35c3b95e43c193e97834a0cedf5e232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:22:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95b847d690d74c6593c5c18906522f1f : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725bb25580284ec79564133b855a7fa5 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e8cc52da4c94481819154221c8ff302 : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1087e2ed5a2d42bcb4764e6b5a1c0de1 : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bedbb50bf8bf4872bac6931b849e62b8 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09523141d4d14ff8bac0d87405137398 : SUCCESS in 11m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/70d9dd39bd564d8a89bc6177278c25d7 : SUCCESS in 1h 12m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3381eb0d71a140478e4d4b9c2a3c11e9 : SUCCESS in 1h 25m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/700fc864484d480bbb365dc11c0bc23e : SUCCESS in 7m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f113ea2d98124348b37647a9fb9ee55a : SUCCESS in 13m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3bbfc0cfe73d4428ac92ce007be4c872 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef11ca3389ce40e8a538612bc166b782 : SUCCESS in 6m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/112759d9d54946169d725a0ca8fba1a4 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee440177f01748f8bfd52af295e800d5 : SUCCESS in 40m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8392d4328fff4f1b8db9168ce67dc40d : FAILURE in 7m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/85d54081c6304f08ae074f34e8e00dd7 : SUCCESS in 1h 20m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/541b395de7724b8ea0f89f4657894fdf : SUCCESS in 1h 20m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/38bc3066393747a89d4b348390506eaa : FAILURE in 1h 53m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4835ac5ab7d64fcf8a7f159f290f5458 : SUCCESS in 1h 27m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d5cd4af823c49e6b3c6600c6cd295ac : SUCCESS in 1h 25m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7016b92bcb22492eb1a7b6b6de22a0eb : SUCCESS in 1h 19m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dad17ef4cd62446e8a79a5dc18c72ba2 : SUCCESS in 2h 06m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/149cc85989294af8b0d8d53452b04311 : FAILURE in 1h 01m 21s","accounts_in_message":[],"_revision_number":9},{"id":"6a402bb6cd5b2becb9f1587a3a93d3b0ad3c81d2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 18:37:50.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/782405/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/164eb35/164eb35.html : SUCCESS in 2h 20m 05s","accounts_in_message":[],"_revision_number":9},{"id":"34be17fdf9f2d824978a1358e35e31143057b05e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-19 21:17:37.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_nfs : SUCCESS in 4h 36m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_iscsi : SUCCESS in 4h 42m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"e9321ca171c1849d5e066c07be82dc413237c8a6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-20 02:57:35.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_nfs : SUCCESS in 4h 52m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_iscsi : SUCCESS in 5h 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"3d6fcc24810bad682c3128e71d156babf09d730a","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-05-20 03:40:12.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"1fee5b44fc0df7402337d42f754b59e4265d982f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-20 03:49:36.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/782405/9/check/cinder-storpool-tempest/93b2a2a/ : FAILURE in 7m 34s","accounts_in_message":[],"_revision_number":9},{"id":"60d22ea756be43a3e59982441b1204bb31f0816c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 04:04:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17d8ce47219b445796f6ce8a2cc301ad : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28949e9bdcc84705a773301e1697846b : SUCCESS in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"77a69021865ebe751314fc48525e3ddf2756798e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-20 04:37:45.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/9/iscsi/ : SUCCESS in 56m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/9/smb/ : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":9},{"id":"7557a1bf11c2d37e5ac00c879d42f216eee44f08","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-20 05:01:46.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/9/check/dsvm-tempest-cisco-zonemanager-job/6520788/ : FAILURE in 1h 16m 16s","accounts_in_message":[],"_revision_number":9},{"id":"f66e4797c5e59607248426cddf8d69a6841ee925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 05:51:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59494742ff54085819f49110fb9daa5 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86ff308f2d54a1a949f14b07b7a726c : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d345053d4c1a43178ff8274a150fb970 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/178d10bd0e69488ea1fe56664d27d44b : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7d2753487b2444d864a04f53e4ce594 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94788f1010b9443db33db196dda2ba9c : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/5a232e18be25401bb9bb6a074027c90a : SUCCESS in 1h 12m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59814dc0044d4ebd9c56f1edd69fa1a9 : SUCCESS in 1h 08m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/960a2dea6e6642b8b4d3eefdf945baf2 : SUCCESS in 7m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc74ea042f54444fac6bca2f5febf994 : SUCCESS in 13m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48a4a8fe7ad84786ae9cf03893adb957 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c60c2b5a81dc4154a6d9ea96680d720a : SUCCESS in 6m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a520404dddbb47ff9b5ae223001c1ca2 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1047f226cbf4ca9a7cb9db5daa3e29b : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01d28eb8a96f4196951038fce04ce2af : FAILURE in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f9f222b63db841b9826926aa0e646340 : SUCCESS in 1h 14m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/075691053d4b441bb0aa1a1757a8f4da : SUCCESS in 1h 17m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e052896ed748426fa163c7a2e73aae81 : SUCCESS in 1h 35m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d116977c1d7407c86e74e102ad4b9ee : SUCCESS in 1h 17m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/89ffc021f40c46e4992fe337f4601f9a : SUCCESS in 1h 18m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f5475597fc7341368838e7170d4892ea : POST_FAILURE in 39m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d2c99662e43747d581c971b1ab43b0f6 : SUCCESS in 2h 05m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a18e76855804b88bbd1e13295d5228b : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":9},{"id":"c7358312c71cbdb87a2144643208ea1e22f15da4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-20 06:22:05.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/9/upstream-check/cinder-cDOT-iSCSI/659e6da/ : FAILURE in 28m 44s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/9/upstream-check/cinder-cDOT-NFS/285a578/ : FAILURE in 29m 40s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/9/upstream-check/cinder-cDOT-FCP/470dd39/ : FAILURE in 14m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"8c35644ad7237acd18ec3215682eca19c417a4ac","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-20 06:22:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/9/upstream-solidfire-check/cinder-solidfire-iSCSI/11dcd69/ : FAILURE in 29m 42s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"29d587d0e0886e53684b5b1ffc7ea2956f4414e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-20 06:25:05.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/9/check/tempest-dsvm-full-focal-py3/880a5d0/ : SUCCESS in 1h 57m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/9/check/devstack-plugin-ceph-dsvm-focal-py3/ed4f001/ : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":9},{"id":"d689b51fdc649a24d82267bfa933f92322130a7d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-20 06:45:04.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/782405/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/570833b/ : TIMED_OUT in 3h 02m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/782405/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/64a87a8/ : TIMED_OUT in 3h 03m 46s","accounts_in_message":[],"_revision_number":9},{"id":"3c910ded666d766e7886c9d3bbe770c966ecbd88","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-20 07:08:05.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/05/782405/9/check/DellEMC_PowerFlex/920b173/DellEMC_PowerFlex/112 : FAILURE in 1h 16m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"4f259d0e76479c59b8f6a621fe6ce2ce67712bc7","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-20 07:46:58.000000000","message":"Patch Set 9:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/782405-9/Seagate-iSCSI : SUCCESS in 45:37.59\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/782405-9/Seagate-FC : SUCCESS in 39:18.03","accounts_in_message":[],"_revision_number":9},{"id":"53a267a8e97c3c543d5114aa45bef30baed6851e","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-20 07:58:24.000000000","message":"Patch Set 9:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/782405-9/Lenovo-iSCSI : SUCCESS in 50:27.46\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/782405-9/Lenovo-FC : SUCCESS in 47:05.86","accounts_in_message":[],"_revision_number":9},{"id":"a27cfe2ff7815f1c793104e7a1a1a9a839ed33e0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-20 08:21:52.000000000","message":"Patch Set 9:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/782405-9/HPMSA-iSCSI : FAILURE in 1:27:12\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/782405-9/HPMSA-FC : SUCCESS in 34:02.61","accounts_in_message":[],"_revision_number":9},{"id":"f7ac09260791c6382ede0a4c6bd89e50e892a16e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-23 04:44:21.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/9/20210523110811/NEC-ISCSI : SUCCESS in 2h 33m 07s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/9/20210523110811/NEC-FC : SUCCESS in 2h 36m 06s","accounts_in_message":[],"_revision_number":9},{"id":"fcf0e6f5268dbbaa2131c10585dc36a50bc7a17d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-30 13:33:30.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_nfs : FAILURE in 6h 26m 16s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_iscsi : FAILURE in 6h 25m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"5a2dbf15c04b8fc4a0cb73b1ca735f05d147441c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-31 10:53:03.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_nfs : SUCCESS in 5h 1m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-9/ns5_iscsi : SUCCESS in 5h 8m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"e85f32e9a6ca32c27e82f4b2a37cee59fee1820d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-06-01 10:04:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cc1184570f3b6a613ec65768c4f1d7cd2769d038","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-01 10:16:52.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"d0ac192fa854882f7f9f941dc95163559209e435","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:38:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a03b3d750da54d2ca12596b9b84bdad4 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c11fc2a788414f70ae248390975a1752 : SUCCESS in 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"13f20f03dc5d9e8cb535ec2fdcbc309df8d6e96a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-01 11:01:30.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-primera iscsi-driver-primera : MERGER_FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":10},{"id":"030fa4b4685de767f28d5a83401ea95efc01293b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-01 11:19:51.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/10/iscsi/ : SUCCESS in 56m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/10/smb/ : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":10},{"id":"5220b054cc308d7fb63a1de7493efe42cc475101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 11:59:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fde43d6e2064f2f94f67105c297ca24 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7faece3702741acb5ab7e89ae4d2968 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9605642651284300abefcce5b66a4e26 : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88fa1f04ba4d4e9da4a036bd10abcd27 : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e6e89c7ffb4d079f5471f370cf6fc4 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb9894bd5bc433cb39bdb7b2f56e8a3 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/da99859bb9584d71973562f050239d6a : SUCCESS in 1h 02m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdcb61d1aeb9410ca8c3f657bc3346ab : FAILURE in 1h 00m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55778e56d9b84a2982967568c8409525 : SUCCESS in 6m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7527d560dcb64f12959c617b6305d366 : SUCCESS in 12m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08e5659f07c7470da641a5da14ab3f1a : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bc13ae088a244d280fbfa918ec20132 : SUCCESS in 8m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dab04db3cd7f4221989e50d12377855b : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/619b757b238f4569a813483bc5694569 : FAILURE in 29m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c31e166c3ee04a47bd559c12fc71bdc4 : FAILURE in 9m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d4a2a2ee4bd4b6bba7589f73fbe27a2 : SUCCESS in 1h 33m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d766860a6ef46a29523ba437a1a71dd : SUCCESS in 1h 25m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/822aff5af60f46388b5814f4cb14f238 : SUCCESS in 1h 19m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/731bb1fd17724d5897edc665162c48ef : SUCCESS in 1h 08m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/74cf6a41f4ed4a5f85b5a3b8e31a4e4f : SUCCESS in 1h 05m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f57a31276c341b29dc632cf18e30675 : SUCCESS in 1h 05m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ef8930439194cb4bdba9f0a8f5472a8 : SUCCESS in 1h 53m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a000ecb8d8543088c40c5e53e69bf2d : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":10},{"id":"f06b4d91fff1019a56895ee3e6121f07958cf07e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-01 12:33:01.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":10},{"id":"38014c366fd657ee977f11fd7e9f67a388607906","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-01 12:40:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/10/check/dsvm-tempest-cisco-zonemanager-job/5412995/ : FAILURE in 58m 37s","accounts_in_message":[],"_revision_number":10},{"id":"c5eee3701f1d7e6e9e50e35144a4bd7af48842fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-01 13:02:25.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/10/check/tempest-dsvm-full-focal-py3/9570ee8/ : SUCCESS in 2h 20m 38s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/10/check/devstack-plugin-ceph-dsvm-focal-py3/3415165/ : FAILURE in 43m 28s","accounts_in_message":[],"_revision_number":10},{"id":"ead243cae5bfe0c64ee73dbc2b7625234c0d3eda","author":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"date":"2021-06-01 14:00:44.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a2bb4966391cdc34d62aea1db7df76b9bc90f970","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:25:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f105b4080f246d0b4376e978124f78a : SUCCESS in 17m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0434cb124dee4de89a498498e507ccb5 : SUCCESS in 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cf0d9c9c37f10085ee6e5629873c837a9ea69635","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-01 14:55:20.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/782405/10/iscsi/ : SUCCESS in 53m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/782405/10/smb/ : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":10},{"id":"7b9b572869549af2e602ae2a7256cfdb3569f2ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:19:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4a0440f25be486d8fbe3357c57ba255 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cfb1d36cf542e6aa2d9381766b9779 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f3143bc19884c47ba95beb2419c1555 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92cae25334b64c97843bc9a024da5a7e : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9e84bcb2584453d9fcce14dce027024 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb7e98f58574230994585d846baff42 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ca0e88842d46461da30b33bff29dc58f : SUCCESS in 1h 12m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32d60550da8848f28a43654ee1e6e2f5 : SUCCESS in 1h 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/976ce99a5a1042eb816f710e42f45dcf : SUCCESS in 6m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/15c9371aff4e4547b267564983019e9f : SUCCESS in 20m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78da92dc4cd541dcb7968904dd7447cb : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf9db195b5d54e16b65a232ae0c8e2c3 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56bc40c123684c31b9702237612792f8 : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f2e0d4725259459d8b7942d42dcbf53c : FAILURE in 25m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fa7a5da91c449248c7b46148c5539e6 : FAILURE in 9m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2f7bd48f7d8a4659b952f2d026a6b498 : SUCCESS in 1h 22m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6ff8c67828ff4bc497a1257afd00a5c3 : SUCCESS in 1h 37m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea6615a8d23241a7bd91aee9add0bb18 : SUCCESS in 1h 32m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00b2451424344abba08bef1f8f8b97e0 : SUCCESS in 1h 00m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9acefad1ac414e0487d2e08d376f7a25 : FAILURE in 1h 29m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/287a446d82b849f8b32504e5f5a06c0e : SUCCESS in 1h 35m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6df054aaa0e2401e925f64affb56667f : SUCCESS in 2h 15m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/832df09e10544606bd39d7c02c19ac39 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":10},{"id":"0d8cea68d95449d94bb5e040f9ccbeda63385f2c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-01 16:20:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/782405/10/upstream-solidfire-check/cinder-solidfire-iSCSI/803fea9/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"f1e5bda5262b801c3811d56b453b99a248116575","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-01 16:21:54.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/782405/10/upstream-check/cinder-cDOT-iSCSI/acf36f7/ : POST_FAILURE in 1m 40s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/782405/10/upstream-check/cinder-cDOT-NFS/6883d37/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/782405/10/upstream-check/cinder-cDOT-FCP/2f7200e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"050045717ee25db72ed6af1b1db0f9f23ce3f2a5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-01 17:01:20.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/10/check/tempest-dsvm-full-focal-py3/aedec66/ : SUCCESS in 2h 20m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/782405/10/check/devstack-plugin-ceph-dsvm-focal-py3/95e2323/ : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":10},{"id":"d2d073da8c3968bc4e061e0a908280645e7f0e9b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-01 17:01:28.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":10},{"id":"dc5eb08bbe0296bf29db2b5b51d5d32dc5e8f73a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-01 17:04:04.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/782405/10/check/dsvm-tempest-cisco-zonemanager-job/4f6b249/ : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":10},{"id":"6101afdaef7f42aa283628418d50f22274678987","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-01 19:09:03.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-10/ns5_nfs : SUCCESS in 5h 9m 27s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-782405-10/ns5_iscsi : SUCCESS in 5h 15m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"fe08271b04bcc85ad02fada8606d25c89bfeeb73","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-06-02 06:00:41.000000000","message":"Patch Set 10:\n\nrun-IBM Storage CI","accounts_in_message":[],"_revision_number":10},{"id":"314f2698ffc49de3c354de02b1e1df4f92654a50","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2021-06-02 06:13:00.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/2101/index.html : FAILURE in 10m 28s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/2247/index.html : FAILURE in 11m 40s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/2236/index.html : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":10},{"id":"5561084444ed4310d1a9d1593a4c42be60ff23d7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-02 09:32:28.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/10/20210602154538/NEC-FC : SUCCESS in 2h 23m 58s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/782405/10/20210602154538/NEC-ISCSI : SUCCESS in 2h 46m 47s","accounts_in_message":[],"_revision_number":10},{"id":"958cba6675379c01429dd33069b3b80f84d2280b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-08 20:36:55.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/782405/10/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":10},{"id":"72d739a893996cc738cf4945811805d3697cdf85","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-21 09:30:42.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"23b87537dd55b96fa3fdf9b359e6ea604f391926","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-07-23 05:15:20.000000000","message":"Patch Set 10:\n\nrun-IBM Storage CI","accounts_in_message":[],"_revision_number":10},{"id":"861c1c2181e344f753a8006864a83f22e79642e1","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-07-23 17:55:21.000000000","message":"Patch Set 10:\n\nrun-IBM Storage CI","accounts_in_message":[],"_revision_number":10},{"id":"bbea138ead26f21c7d8a75946122b3b67fd69220","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-07-23 20:07:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/3427da2/index.html : SUCCESS in 1h 05m 05s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/bd29557/index.html : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":10},{"id":"a02066d503c479a86911a1e0f642c63d294c7644","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-07-27 11:29:09.000000000","message":"Patch Set 10: Code-Review+1\n\nCI is green, UT is covered. LGTM","accounts_in_message":[],"_revision_number":10},{"id":"3fd29657b9f178c8674c4603ece50ba1337b435c","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-30 21:12:52.000000000","message":"Patch Set 10: Code-Review+1\n\nI\u0027m not familiar with Storwize so I need to debug this a bit more. In the mind time I\u0027d like to see another reviews.","accounts_in_message":[],"_revision_number":10},{"id":"e994f03f93445515cf165f30dfe59cf7377ec998","author":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"date":"2021-08-02 12:48:14.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"5e6d514711e779869151a3181b26e53ca0c87d9e","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-08-03 12:44:44.000000000","message":"Patch Set 10: Code-Review+2\n\nPassing CI.  Change looks ok.","accounts_in_message":[],"_revision_number":10},{"id":"b453e7209e0f756fe99370836e1ab16942c167cf","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-05 16:44:14.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"434a4aef0103ef77711709b1d42aa55c11aee73e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:44:50.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"23617015aabb3d55da12386c26e0174bb6421f4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:12:02.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5007ced32014ce2991e668418bda8e4 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6285ea81008c4dc798cc17f759c47851 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7f64f2ce66c4d38ba89dc71fe5ccc38 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee091b307ea84dd88643bdedb28f5d83 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/dc1d0da0627a4fdab3fcc575a0d3b587 : SUCCESS in 58m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/404a65b6d4844798a49e18f2f9c54c05 : SUCCESS in 1h 03m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e08fa99e5dd940939f907b578223e60e : SUCCESS in 7m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27f1e8b99db84a01840be8b81924cb71 : SUCCESS in 1h 09m 24s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1fb4eaba898b476cb396b88bd2f1bfb1 : SUCCESS in 1h 08m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e7fc648acac4302823e65ca5e2d9f62 : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":10},{"id":"70416027c58d6b62de586f9773a7192157166835","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:12:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"456c8dfd935448edf831d44fe57593ed6e722a26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:16:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d636c6618b94586b15fcb8b67898ea5 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341d0b377fe84d4ebcd67efde5a589bd : SUCCESS in 57s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"020944bada1f00e0180d8da6fedc7236ba6445b6","revisions":{"9966e7550f56092104dc4ccfb35d229750500393":{"kind":"REWORK","_number":1,"created":"2021-03-23 10:48:25.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/1"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:47:59.000000000","tz":330},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy mapping\nbetween clone volume and its source volume was not being deleted after\ncopying, hence extend operation is failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9966e7550f56092104dc4ccfb35d229750500393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9966e7550f56092104dc4ccfb35d229750500393"}]},"branch":"refs/heads/master"},"a8ce54a3d0599c95df623151c2cfe657b697fb8d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-23 10:58:26.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/2"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:53:00.000000000","tz":330},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy mapping\nbetween clone volume and its source volume having snapshots, was not being\ndeleted after copying, hence extend operation is failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-bug: https://bugs.launchpad.net/cinder/+bug/1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8ce54a3d0599c95df623151c2cfe657b697fb8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8ce54a3d0599c95df623151c2cfe657b697fb8d"}]},"branch":"refs/heads/master"},"8e243399cc74b61e309342e6aea5041292ad882b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-23 16:03:20.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/3"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"Amar","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 16:03:20.000000000","tz":0},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation is failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-bug: https://bugs.launchpad.net/cinder/+bug/1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e243399cc74b61e309342e6aea5041292ad882b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e243399cc74b61e309342e6aea5041292ad882b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0a1620d16a516d6cc08de9255c0a29ea0fe8a1cd":{"kind":"REWORK","_number":4,"created":"2021-03-24 16:32:52.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/4"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-24 16:29:04.000000000","tz":330},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy mapping\nbetween clone volume and its source volume having snapshots, was not being\ndeleted after copying, hence extend operation is failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-bug: https://bugs.launchpad.net/cinder/+bug/1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1620d16a516d6cc08de9255c0a29ea0fe8a1cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1620d16a516d6cc08de9255c0a29ea0fe8a1cd"}]},"branch":"refs/heads/master"},"cc52650eb70e2bba9d6769b37dfc2b35d8973b42":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-04-05 17:57:59.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/5"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-04-05 17:52:39.000000000","tz":330},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation\nis failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-Bug: #1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc52650eb70e2bba9d6769b37dfc2b35d8973b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc52650eb70e2bba9d6769b37dfc2b35d8973b42"}]},"branch":"refs/heads/master"},"f4cc69561b3e62785255fe1a15b7c4d8bd013a2b":{"kind":"REWORK","_number":6,"created":"2021-05-05 12:09:51.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/6"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-05-05 12:09:08.000000000","tz":0},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation\nis failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-Bug: #1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4cc69561b3e62785255fe1a15b7c4d8bd013a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4cc69561b3e62785255fe1a15b7c4d8bd013a2b"}]},"branch":"refs/heads/master"},"4d1fff8c3cbabd81b420bb98cce033a1ba8090d3":{"kind":"REWORK","_number":7,"created":"2021-05-05 13:10:17.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/7"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-05-05 13:09:31.000000000","tz":0},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation\nis failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-Bug: #1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d1fff8c3cbabd81b420bb98cce033a1ba8090d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d1fff8c3cbabd81b420bb98cce033a1ba8090d3"}]},"branch":"refs/heads/master"},"3320670d45aa05d876ca878031e00668ed41d795":{"kind":"REWORK","_number":8,"created":"2021-05-19 05:06:01.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/8"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-05-19 02:23:28.000000000","tz":0},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation\nis failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-Bug: #1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3320670d45aa05d876ca878031e00668ed41d795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3320670d45aa05d876ca878031e00668ed41d795"}]},"branch":"refs/heads/master"},"b7d3a2dedd8c2839e40d474802bacd568dfd3a66":{"kind":"REWORK","_number":9,"created":"2021-05-19 16:15:32.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/9"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-05-19 16:12:56.000000000","tz":0},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation\nis failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-Bug: #1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7d3a2dedd8c2839e40d474802bacd568dfd3a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7d3a2dedd8c2839e40d474802bacd568dfd3a66"}]},"branch":"refs/heads/master"},"020944bada1f00e0180d8da6fedc7236ba6445b6":{"kind":"REWORK","_number":10,"created":"2021-06-01 10:04:31.000000000","uploader":{"_account_id":32075,"name":"Amar","display_name":"Amar Rajeev Singuluri","email":"amar.rajeev.singuluri@ibm.com","username":"amar7ibm"},"ref":"refs/changes/05/782405/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/782405/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/782405/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/782405/10"}}},"commit":{"parents":[{"commit":"a01b5bb39e9d2b6163b30f592f78ffe7e26256a7","subject":"Merge \"PowerMax Driver - Allowing for all types of boolean in extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a01b5bb39e9d2b6163b30f592f78ffe7e26256a7"}]}],"author":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-03-23 10:41:41.000000000","tz":330},"committer":{"name":"amar7ibm","email":"amar.rajeev.singuluri@ibm.com","date":"2021-06-01 10:03:51.000000000","tz":0},"subject":"[SVF]: Fix extend issue for a clone of rep-volume","message":"[SVF]: Fix extend issue for a clone of rep-volume\n\n[Spectrum Virtualize Family] During extend operation the flashcopy\nmapping between clone volume and its source volume having snapshots,\nwas not being deleted after copying, hence extend operation\nis failing.\n\nThis fix is to properly stop the flashcopy mappings before extending.\n\nCloses-Bug: #1920099\n\nChange-Id: I336a8c92bfdd852cab42c9e5f992a30f3632bfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/020944bada1f00e0180d8da6fedc7236ba6445b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/020944bada1f00e0180d8da6fedc7236ba6445b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
