)]}'
{"id":"openstack%2Fcinder~782917","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~I5e35de28007558267f9215cbac03a38d7c74612c","project":"openstack/cinder","branch":"stable/victoria","hashtags":[],"change_id":"I5e35de28007558267f9215cbac03a38d7c74612c","subject":"Add support for RBD fast-diff feature for backups stored in Ceph","status":"MERGED","created":"2021-03-25 08:44:13.000000000","updated":"2021-04-27 14:52:22.000000000","submitted":"2021-04-27 14:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782917","meta_rev_id":"51268c80d34118bcff43def29959ec631b7de034","_number":782917,"virtual_id_number":782917,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 14:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-04-16 07:23:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-04-14 14:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-14 14:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-27 12:56:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 08:46:11.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-25 09:23:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2021-03-25 09:23:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-25 09:23:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-25 09:23:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-03-25 09:23:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2021-03-25 09:23:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-03-25 10:12:53.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-25 10:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 12:25:00.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-25 14:53:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-25 18:15:38.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-25 18:48:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 08:17:20.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-27 08:09:48.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 14:27:23.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-04-16 07:23:20.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"8562dc8d34a74fec9b3d7787ccd69f3a325fec59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-25 08:44:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e50cd32891d3ce106d448f8f341480afa2f38c82","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-25 08:46:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9be529d1252be5dafb427062e4dfce8eeedc5019","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-25 09:23:28.000000000","message":"Patch Set 1:\n\nI cherry-picked the now merged https://review.opendev.org/c/openstack/cinder/+/766856.\nSince you were the reviewers of the other change I took the liberty to assign you.","accounts_in_message":[],"_revision_number":1},{"id":"3ad872d5d715b635381ed1e19ee20f0e535f53d5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-25 10:12:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/17/782917/1/check/dsvm-tempest-cisco-zonemanager-job/ff813ba/ : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"82325f7a4b38963a8312bf78f8dd58a331140964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 10:53:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfbedb8d055744ffa0bd3bd6d0b63a11 : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a22997fe2f64b9b8dab1a2ac637eaf1 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/308be0ee5caf4804b02526f7d1c03831 : SUCCESS in 12m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dce6f9b83d74c64be6e6f3bc621001b : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9035a40432bc4b039b57695b19a20e9a : SUCCESS in 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/229bf3e387614f75b65ad048e0ab020f : SUCCESS in 56m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01639bea31e34a05816a56fb2cb585db : SUCCESS in 1h 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37b8ff7e04e4a4482b98f063a81cb88 : SUCCESS in 7m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f56e52a615df44ff988b46a516187e4e : SUCCESS in 13m 32s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22e7f74f733b44eba867aff350640053 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a171ba927c2d46708eb93f543ac6fc41 : SUCCESS in 6m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc7b88f72203419db28a827ec89c9063 : SUCCESS in 9m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/725842de4a8a4fbbbbf9b374042af22a : SUCCESS in 43m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ef5c0b42a5242b49da0754e658bc57d : FAILURE in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efa8b647f8a14d118222ffd86b1c6daa : SUCCESS in 1h 15m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a1b7c78d92a84c0f809aada45e55ae17 : FAILURE in 1h 41m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c3940d047fc43d28c50fa453d99b898 : SUCCESS in 1h 34m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d76846871e0548c99432f360571ec3da : SUCCESS in 1h 14m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b7441c673db479f9dc3737d3c663f25 : SUCCESS in 1h 12m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b7a8dd5bb9764f40ab1037eaf4adcdf3 : SUCCESS in 1h 22m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7f52678ec204ff68626d24957c6e4f1 : SUCCESS in 2h 03m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f1b990ceb1f47b88a68fd4a0eb5ea8d : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"51eb017a53aa71a8499549f44f74ac52c816711a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-25 12:25:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/782917/1/upstream-solidfire-check/cinder-solidfire-iSCSI/fc6e5e5/ : FAILURE in 1h 29m 29s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"c5b295194ccd0d1e2297273228de6af3714ad622","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 14:53:22.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-782917-1/ns5_nfs : FAILURE in 5h 32m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-782917-1/ns5_iscsi : SUCCESS in 5h 7m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"38a25c00a4c6f5577d71f0df95c46c005c4ae091","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-25 18:15:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/17/782917/1/check/dsvm-tempest-zadara-driver/b2edeb0 : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"526a7c33d31fd93b87d8d231cef404f243271fcd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-25 18:48:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/782917/1/check/iscsi-driver-nimble/7d4ddf1/7d4ddf1.html : FAILURE in 1h 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d699a8aad530c854e9c83c718382ac6f8fec5221","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-26 02:10:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-782917-1/ns5_nfs : SUCCESS in 5h 9m 46s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-782917-1/ns5_iscsi : SUCCESS in 4h 59m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ad06044c0868062b8fbed1090e59349722d744fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-26 08:17:20.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/782917/1/check/tempest-dsvm-full-focal-py3/2db7852/ : TIMED_OUT in 3h 09m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/782917/1/check/devstack-plugin-ceph-dsvm-focal-py3/8066fef/ : FAILURE in 1h 30m 41s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud finger://zuul-v3-executor.novalocal/4b49716587c74431811237b18a7b09fc : POST_FAILURE in 1h 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8e89e35461550db82b2ce5840f113931a3c15b44","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-27 08:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/782917/1/20210327130622/NEC-ISCSI : SUCCESS in 2h 23m 03s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/782917/1/20210327130622/NEC-FC : SUCCESS in 4h 03m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2632d9f490c631a9d16f119195008030db94791b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-04-14 14:27:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b3becda341f03574b42b14a127d7e00eb65cbfad","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-14 14:28:06.000000000","message":"Patch Set 1: Code-Review+2\n\nTeam is ok with backporting this change so I support.","accounts_in_message":[],"_revision_number":1},{"id":"a3b58c08c38a5bca14c55340d6e3fce7787a7a53","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-04-15 10:42:16.000000000","message":"Patch Set 1:\n\nWhat else is required to have this merged?\n\nI\u0027d like to propose backporting this to ussuri (https://review.opendev.org/c/openstack/cinder/+/782929) and there even is a change by Andrey Bolgov for a backport to train (https://review.opendev.org/c/openstack/cinder/+/784041)","accounts_in_message":[],"_revision_number":1},{"id":"f53f2ffcdab080b0358279b7424eb85626b8b23f","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-04-16 07:23:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"58effd3c9ba2b0f04c63d7c25d2b32c933d9923f","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-04-26 12:51:54.000000000","message":"Patch Set 1:\n\nThanks folks for the +1 / +2. What is missing to have this merged now?\nWho does trigger that?","accounts_in_message":[],"_revision_number":1},{"id":"ea501e2e84e90a57d6f7fa7411fc473c8fecb709","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-27 12:56:53.000000000","message":"Patch Set 1: Workflow+1\n\nTwo core reviews so at this point it should be safe to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"fcc26948cb661703593b0c3c213e61298dca6745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dff300d64aad83a4e677f70f42ec493b4cb99224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 14:43:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a2b90ac6034e3abbabe5fddc2dca8e : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebef1cde83e4468497745975e2e63134 : SUCCESS in 15m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c99b534fb654e61850029c26972f838 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9084f70c071e470da801b7247bcdf54d : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/092e38e7b12d46aa9b05d0fc4d95f23a : SUCCESS in 46m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da918c42588349f6b641ad0668e16019 : SUCCESS in 1h 08m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aaccefb8d4943ddb8401550843dd6b2 : SUCCESS in 7m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/96f3b1cf765b42a6a86450d416f879da : SUCCESS in 1h 16m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a45e7d70a842458e9b8f91ddf32baa38 : SUCCESS in 1h 25m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/779085a108e741c4946d9baf89a0f2cf : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d8c6078c284937b79faca1d9f6b4ad4f74c65f04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 14:43:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51268c80d34118bcff43def29959ec631b7de034","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 14:52:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74df1afda4904bd3a7e28da3479099e7 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6d729b98c814fd3b2dc0dd41ca006de : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbf9def32b31bcc3d4894d05e6fe665005006492","revisions":{"cbf9def32b31bcc3d4894d05e6fe665005006492":{"kind":"REWORK","_number":1,"created":"2021-03-25 08:44:13.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/17/782917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/782917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/782917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/782917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/782917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/782917/1"}}},"commit":{"parents":[{"commit":"990cba70eebebf2af484b00cb3771655c979076d","subject":"Merge \"Label temporary files created by image_utils\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/990cba70eebebf2af484b00cb3771655c979076d"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-25 08:43:46.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature for backup images stored\n    via Ceph RBD. It is enabled together with EXCLUSIVE_LOCK and\n    OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is sped up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n(cherry picked from commit 0e8807388209086ee60e1c9f11a48ed6191e4178)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbf9def32b31bcc3d4894d05e6fe665005006492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbf9def32b31bcc3d4894d05e6fe665005006492"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
