)]}'
{"id":"openstack%2Fcinder~783204","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I1349e8d3eb422f9dcd533c54f922f7ab8133b753","project":"openstack/cinder","branch":"stable/ussuri","hashtags":[],"change_id":"I1349e8d3eb422f9dcd533c54f922f7ab8133b753","subject":"API validation: Add cinder_host type to support ipv6 in manage","status":"MERGED","created":"2021-03-26 13:33:44.000000000","updated":"2021-05-05 17:10:48.000000000","submitted":"2021-05-05 17:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783204","cherry_pick_of_change":780049,"cherry_pick_of_patch_set":1,"meta_rev_id":"282d37a4d28ecc22e444e81ad281f5302862774a","_number":783204,"virtual_id_number":783204,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 17:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-05-05 14:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-04-12 15:37:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-13 15:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-05-05 14:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 13:35:27.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-26 16:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-27 01:51:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-27 04:03:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-27 11:17:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-27 12:07:33.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-28 06:43:10.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 10:15:05.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-03-30 19:53:19.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-07 12:23:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-04-07 16:39:12.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-12 15:37:00.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-04-12 17:03:15.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-12 17:21:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 15:48:26.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-05-05 14:35:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"08e6d756f0d5e19659e341e440ac2e0f7e500c7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-26 13:33:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"7a3b1373f657aa66102c8e96c4415a085bfff182","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-26 13:35:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4abd8a710ed93adb765f3afddbd5758809edac38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:09:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67f8355ea03248af842128d63325e84c : FAILURE in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19188586e48b424bbfcbf693cf85badb : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e840cecce9c4cff884770e1fad9b33b : SUCCESS in 9m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9260701cacd74768896bf47d27c25999 : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92701eda60443189eb64e01e397d292 : SUCCESS in 18m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/681a166dc76d426382cbc747169270e0 : SUCCESS in 1h 09m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2aa7a1b971af4123800f612e9c3bc8a7 : SUCCESS in 1h 05m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b2874963be94a15bccea02f7a5d92ba : SUCCESS in 6m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7bbf5607016243c9a2b893f51684fbc4 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/308ff37e07fa473b8870b037c612afac : SUCCESS in 8m 13s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/540a6bfd3c5d47f1bc61289cee53e7c2 : SUCCESS in 9m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b163e8d467a744019eaeb158a066a743 : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1bd3e5f57f584ebfbbd8af40965c4db2 : SUCCESS in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11b9479a9b9c4126969064b48e93fdeb : FAILURE in 1h 23m 09s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c2d574c0a74843129ddc049578e0855a : SUCCESS in 1h 32m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1554841fcfe84075b2dfbb4e6bf4a9fa : SUCCESS in 1h 26m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d809e2e0119e44d5bb040135d41ee1b4 : SUCCESS in 1h 37m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34336a9864a14196a0bddd07963390aa : FAILURE in 1h 16m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a53f7bbf1e2413f81c8d0f2b6a03c5a : SUCCESS in 2h 32m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07693c635d844af4a26d87aad16df30b : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"712b3217f71e83eb8a804e6ade8b664f0d8f8a6d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-26 20:56:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"52f7a2078e243a132cc52778c03ef4ca1c88143f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 23:06:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f87aee599fb1487dbb6de66970b2c512 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de6585b0585415595e1387a2847a17e : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11c2a0e4f114888b23f1d54c1a22a62 : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6ce39e4d7184c7d8b3edb7e655e7e0a : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9faab6abcc90466d963082fa52dc8b9a : SUCCESS in 16m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/1b2e802e6df3427391df3cd9d4688483 : SUCCESS in 1h 08m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df365e4f1ead47c6bda163a182fe6cbc : SUCCESS in 1h 14m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bef641d12c741bca0be04d0bd03a09f : SUCCESS in 6m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9eab78fbb9284a2293d845ac0a0a6709 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f72c954ce3a147bda408ebaf8507bb97 : SUCCESS in 5m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5a6f07ab4e234f849857aec879a27772 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f12334465a994fb4b25587ba84df0b95 : SUCCESS in 40m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/992d0a91fa404eeda69bb7222e8f3d97 : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/442f6f1c222944058f3381c5c79918e8 : FAILURE in 1h 17m 32s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/757c4f89498b42498fe8b80600a7268d : FAILURE in 1h 34m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d39a11caf97408e982d5e9a5a3e91d1 : SUCCESS in 1h 21m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ab0f1887e23241ca84b10dcfdb44010d : SUCCESS in 1h 25m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fb0e39e6ae3344139a5d375904b016ad : FAILURE in 1h 18m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b7d13c55b15546cd96ac6bda7a84219a : SUCCESS in 2h 04m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79847f925a75417ea841360db35e33a9 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f29b520f52b0e1994965b29d45a100283167d878","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-27 01:51:32.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/8985b61dbd22487c94b1b10fbfa383b1 : POST_FAILURE in 1h 13m 51s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/c8be0999b97a44e2aee46a29c3918aef : POST_FAILURE in 1h 12m 32s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/4ff4cd1573b54703a63c93f29b7c15f4 : POST_FAILURE in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"19ae4530e96e7540c9dcba48b58171dcb7fa82c5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-27 04:03:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/04/783204/1/check/dsvm-tempest-cisco-zonemanager-job/ddbf806/ : FAILURE in 26m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e2bae296ed6b81779a22201acdd1806589dda139","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-27 11:17:56.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-04-783204-1/ns5_nfs : SUCCESS in 5h 20m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-04-783204-1/ns5_iscsi : SUCCESS in 5h 5m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8d63caa9fac46e6cdae38fe7a736a8f2551abd79","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-27 12:07:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/04/783204/1/check/dsvm-tempest-zadara-driver/5308cb6 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44b7b678b13d64d51cd951a580103b4721782324","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-28 06:43:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/04/783204/1/check/iscsi-driver-nimble/2631531/2631531.html : SUCCESS in 2h 02m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6e2dbf68f5f9f93f160e3d9bbdf7563ad0d01367","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-29 13:42:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b71fbc50f2de5dbf648ef0078b0e8b3e7b70365e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-29 14:41:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/04/783204/1/check/dsvm-tempest-cisco-zonemanager-job/5ba7ea2/ : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e7637056e897bc1e3ef0f5bcd8418830c0c283b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-29 15:49:41.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/de746dc3d2ad40b898ef23d4c47ec0c6 : POST_FAILURE in 1h 21m 22s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic-py3/48c3b04/ : FAILURE in 1h 15m 50s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/devstack-plugin-ceph-dsvm-bionic-py3/5628292/ : FAILURE in 1h 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"de721400cc4f81f4766a887f19ff3d95112d3c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 15:57:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3b0978952a14f42b3d1bcfe5caadea7 : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee532f4c8e54fbf845137563a06a627 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a82caa46f84054b657107ee68c9a40 : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ef8fdb10aeb42d28c3f0e14a383c0e9 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc3386fabab4a10ba660da523a27a73 : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/44fbb328783a41dda36f497c4f87c3c2 : SUCCESS in 1h 27m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb8c2f2eff154c0483cec0aa46eebed7 : SUCCESS in 1h 27m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52d87cb91c7941deabc099288c324b9c : SUCCESS in 6m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1daeb7319c34b13acb1a4aa054c2d86 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87208dbfee974016b3af56aae4fb95f0 : SUCCESS in 7m 58s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4e4253d5dfa044cdaa63006a317d8c7c : SUCCESS in 7m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/95d976eca4bc4e88a82881a8f6c7c17b : SUCCESS in 42m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4b5658e890949c6873cd0b64b4a1daf : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0bb5d1066c4d4434a09e764ddcbfc0ec : FAILURE in 1h 22m 27s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0dddb9543d84259b872ce63210a68bd : SUCCESS in 2h 01m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/903fba5da643469ba66e3d2cd30fb828 : SUCCESS in 1h 37m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8aa4e0a5ad644a6c902f6672a75a75bd : SUCCESS in 1h 15m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1e96fce0c7d4032956c57ed284e24eb : FAILURE in 1h 13m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/784adc87536a40c4aba6d99a9b7acb37 : SUCCESS in 2h 09m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d77ace21e3fe4364bcd6ba3c4edd93f6 : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f36060b19e3a0cd798c69b3758e491e2de1f3b3a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-29 18:00:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/04/783204/1/check/dsvm-tempest-zadara-driver/3ae78d8 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e72df24f19ff207f83d04e4b292f8372d15500af","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-30 10:15:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6149bd69c9c265ed7b91d7ee68029e09a3ba8cdb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-30 11:55:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/04/783204/1/check/dsvm-tempest-cisco-zonemanager-job/a44397f/ : FAILURE in 26m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d857705b16f3aa488220f480c4798c41a82a23a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 11:56:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81e31a3b19084ec58cd1330c6ac95fd9 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ee902a78764ec9bc02af277b77d906 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/185ee1a818204f3b8245376d324a1ef3 : SUCCESS in 9m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d67ef1d62bd47959e9a0f2e3d8be42c : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a73fe11f550c4d14932d848c952c5aa3 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/3eb20e04cb9f410884bc85d90109390e : SUCCESS in 1h 13m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40634689deb84c0f80173bc8bbf05c0c : SUCCESS in 1h 18m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef6fa68285b94d3aa19ba8dc539ca224 : SUCCESS in 6m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd93d8e0ce00443ca463442c467057f2 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5889548564a94230baf2e34081375142 : SUCCESS in 6m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/08f7fb76c3344c82a35b865e45f282f2 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f63f1af6ab314908b85f80612091f572 : SUCCESS in 37m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c91d743006f54d32ac0ac28166363c47 : SUCCESS in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e33cb6feb7ae4e4a9dcb8e255107a52e : FAILURE in 1h 35m 22s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/82b693f5fdb04a7788dc32a56e6a5821 : SUCCESS in 1h 23m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0b93bd342bd34835958bbbf13854ff09 : SUCCESS in 1h 24m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/000147d7bc4c43968016518aa13daeec : SUCCESS in 1h 21m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1ed292c62a94acbbbfd47a96ffc619d : POST_FAILURE in 1h 19m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/996268989901467a9dfcaee189cc156d : SUCCESS in 1h 35m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f80b3b706312491e9147d58be7926c48 : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3900956e334ba1c144776b1301fbc2eb41efc91e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-30 14:46:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/04/783204/1/check/dsvm-tempest-zadara-driver/0a680fd : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a880c0fe2185325b935738a6aadfcd2fbae6ec2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-30 14:52:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic/b7f3044/ : SUCCESS in 2h 40m 56s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic-py3/1636ba9/ : SUCCESS in 2h 09m 43s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/devstack-plugin-ceph-dsvm-bionic-py3/3ce1472/ : SUCCESS in 2h 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"62b7c23b72220be2cbb1591c8c1c77e952d10fbf","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 19:53:19.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/783204/1/20210331025810/NEC-ISCSI : FAILURE in 1h 55m 02s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/783204/1/20210331025810/NEC-FC : FAILURE in 1h 55m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2c7ab975faf9f6a422203215dd5cf6cdd49f4ae3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-07 12:23:14.000000000","message":"Patch Set 1:\n\nThis has hit https://bugs.launchpad.net/devstack-plugin-ceph/+bug/1921897\n\nEric has a patch up addressing this: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/783880","accounts_in_message":[],"_revision_number":1},{"id":"5dc0352063fdbb1682e111405f9570a3a806101e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-07 14:30:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e05348ba60ea30aead283b60f42b8c4fbe901840","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 16:39:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783204/1/iscsi/ : SUCCESS in 1h 11m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/783204/1/smb/ : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6594e3c9701aae4f54c8b32a7b5cf04e0ba13481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 16:46:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd3dc33a56db4851bae9ab3b5e890d35 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aae0b0c828e43dc997faae793f5efd4 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05b40cefe95c4a248752de5511572e29 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/034a7dbda3cd462a9ca30fdfce99ad80 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306bfda8ce484fd387003e8e56ab6e7f : SUCCESS in 16m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/d31522ef2e51468aa29f1ecb58ca6637 : FAILURE in 1h 11m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b819461b9b9400998a904ffb4b9ae10 : SUCCESS in 1h 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eba2a93b6bf42f09bdb3dde398afa31 : SUCCESS in 7m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7e71fbda80640de8325d5639cae983f : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7943e45db6e342e2aa5e4d4fed1381b9 : SUCCESS in 6m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ca71a96cb2c043058cc60920c1924f8f : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a660761c9096429ca0be37012b92e0e1 : SUCCESS in 47m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/480b270d3c8f4d6d87878d7b21699cdc : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed698b804b3e45dd96ca8ec92c819a23 : FAILURE in 1h 23m 56s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e82ca6170bc43a3b363942f741867bb : SUCCESS in 1h 27m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d1d1ca5c39cc4d70805cd4fe85d27642 : SUCCESS in 1h 27m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bc68577e6e6e455fa5276ec7642ad472 : POST_FAILURE in 1h 30m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3cf02798b81d41faaa9d450428a04548 : FAILURE in 1h 13m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/352fd84fc9b34446a2a450e8309f1720 : SUCCESS in 2h 09m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12666c226fcd40de840c1a19e486f9eb : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bf8711d44211ec8529a72deaa7626cb6e4ba7831","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 16:50:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/04/783204/1/check/dsvm-tempest-cisco-zonemanager-job/5330be1/ : FAILURE in 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c2a9f6230a35daa6d8fdc832a086e033df2ffb50","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-07 17:24:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c015aab3e4b39f07b0fb24d1df57c8dd3752173a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-07 18:25:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic/6b8cd68/ : SUCCESS in 1h 59m 32s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic-py3/87716d6/ : SUCCESS in 2h 17m 50s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/devstack-plugin-ceph-dsvm-bionic-py3/c533f81/ : TIMED_OUT in 2h 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"635831e66dfab574d64aef1649febf960457aed3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 18:54:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783204/1/iscsi/ : FAILURE in 1h 29m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/783204/1/smb/ : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a90ce939abd83ad2be15569dbce2114cceaab200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 19:19:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebcb39b5257d4caeb6fcebe786fb9c3c : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7a533785a249c78e48781fadc8c118 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3147ca688ad744659be460551bb7af3c : SUCCESS in 12m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20f26091d4d54cf49970048481a433bc : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ec65ab5b5e4d429b5c81f22318a3af : SUCCESS in 15m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/154d46abfc4541e19d6a2b9377f84a42 : SUCCESS in 1h 09m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d84af632d89343deab89b9c52e97d8bc : SUCCESS in 1h 08m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9f8c3fafaf44a9d876141baeb14efd6 : SUCCESS in 7m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a04a732929574ed9b28a7789e869c137 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4460274261424eaa8ecf719e4db0defe : SUCCESS in 7m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/293d3b1ad34246998539d793720a4087 : SUCCESS in 12m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/810b9bea3a9b49268fdbd452288281ac : SUCCESS in 38m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7d65081ed5445af8900a9400d01da58 : SUCCESS in 7m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a8a25884d5d44f2ba526b7d4370f69f1 : SUCCESS in 1h 12m 52s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f0cf9cb645f8414ba924066b7dcea5fd : FAILURE in 1h 50m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc5da4831116436d8f1908e0e6effc07 : SUCCESS in 1h 00m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/297bbc0cd86242db87787e361d6e50ba : FAILURE in 1h 24m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/16ca5b29d126403eb7ca413b3d8da683 : FAILURE in 52m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b46cd890caf4b78834483d37ab58018 : FAILURE in 1h 51m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d99b60295ad4b61aaeba9cd5849f240 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c5da2befbd411ea371c6b1590ebcfaa69d6ee7ef","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 20:23:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/04/783204/1/check/dsvm-tempest-cisco-zonemanager-job/c98f114/ : FAILURE in 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5dc2df7eaba34fc25dee4728b7c2817b0fabe8e3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-08 00:32:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/04/783204/1/check/dsvm-tempest-zadara-driver/782380d : SUCCESS in 1h 25m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58aecc6e093f690c454315eea8e3b162a87509d9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-12 13:10:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"42e51327c44514568fec5524eb0dbd070c6f9a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 15:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d04477b2324957a737b663df39146f : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae1d2b12106c4a8e869bcc99733fc158 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75996eb746dc439ea0f9a37f1a54eb58 : SUCCESS in 9m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/049d06deafd54524b9d2064b27d7d919 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f047c564193f48d9a9693e004eab1ef6 : SUCCESS in 15m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/73093be1618e467f89c97052a1eb7a2c : SUCCESS in 1h 01m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4deef3ceb30547a7b9d9dfff194924be : SUCCESS in 1h 14m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db6b9e67caa4718b9a550768a5a0cf1 : SUCCESS in 6m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/26779ef016ad4296bc7bf707b74699e9 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4301b860a0824e2c89d1de206bcda299 : SUCCESS in 5m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5912058affca4c1ebec8f342283d9939 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb8d794dd0104660aeaa5b220587e2e2 : SUCCESS in 38m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bda87b7fb79e4706b6e71ddb5934b962 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4145672ad9646e48ba37bf1266b17f9 : SUCCESS in 1h 17m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/402ba9430a9445ea840dad3a53d941fe : SUCCESS in 1h 38m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/783d0ddfd2ea43859cd3463676ee7df1 : SUCCESS in 1h 21m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b6028ceb77546cbb94647fd673eedbd : SUCCESS in 1h 14m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f727c3fd8f1c4587be3b0a04df4579bf : FAILURE in 1h 13m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b658d4220bb4445966f923c6becd96d : SUCCESS in 2h 11m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff17bcb18ed0482a9bf7f2e3ba285e3e : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"86b5870e84742b22ba435f1ded8c0ea3859364ab","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-12 15:37:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a92486689c518b4347d1d2c90940dad2a63346e0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-12 15:37:26.000000000","message":"Patch Set 1:\n\nussuri is now working, time to merge this!","accounts_in_message":[],"_revision_number":1},{"id":"bdc2de14b00a4f05c83c893f8b5a9161e40691e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-12 16:21:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/04/783204/1/check/dsvm-tempest-cisco-zonemanager-job/fc2609b/ : FAILURE in 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"60c7644fefa825409b0d1bfa3ddacc896a2e57b5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-12 16:24:51.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic/e059628/ : SUCCESS in 1h 44m 59s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/tempest-dsvm-full-bionic-py3/8c8bc38/ : SUCCESS in 2h 29m 26s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/04/783204/1/check/devstack-plugin-ceph-dsvm-bionic-py3/62690da/ : TIMED_OUT in 2h 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ddb4bb8b145dc6b2bddb52685829bfed4090e757","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-12 17:03:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/04/783204/1/upstream-solidfire-check/cinder-solidfire-iSCSI/d26d258/ : FAILURE in 1h 35m 18s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"c80dfa1e6114e77fa65667b5d1a721a055dd87ea","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-12 17:21:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/04/783204/1/upstream-check/cinder-cDOT-iSCSI/4a735ad/ : SUCCESS in 46m 13s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/04/783204/1/upstream-check/cinder-cDOT-NFS/88fe32a/ : FAILURE in 47m 09s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/04/783204/1/upstream-check/cinder-cDOT-FCP/c0b2804/ : SUCCESS in 1h 03m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fa6c03d1ce06dc0b2efa9c7f1dfa7ac62e2f905a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-13 15:48:26.000000000","message":"Patch Set 1: Code-Review+2\n\nI am ok with this backport.","accounts_in_message":[],"_revision_number":1},{"id":"28bb22451b01bcac2f1b6649f343ed028a6a916f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-05 14:35:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is good to backport.","accounts_in_message":[],"_revision_number":1},{"id":"c48ac26b47d5304ee7e2b5d7f37d2551aa241e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5a91e0128e8dcff65dcb339b00500bd60d85a0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:08:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4cb847ea5f4032aff0896c7567589e : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0fb69c2b71442f6a77d3074a51382a7 : SUCCESS in 9m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69dc31c5ba744092a3be8bfe8cba635b : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06d0eb3376741ef8a9ddf82b7365bbc : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/126e5ef9d5d14f129342b9997f260fc3 : SUCCESS in 49m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27428f7cf9cd48538a7498cd6d73afe5 : SUCCESS in 1h 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c1d113c6b84dcfad4730822990c622 : SUCCESS in 6m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7bf2e00d41af4411944d67715e92a5b7 : SUCCESS in 1h 10m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/60140f278bb04f3cb511fdaacfef1d27 : SUCCESS in 1h 29m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0e226af5b164fcda87be6b0d8acb46e : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6e0424335eaa2452291e074738a5e7efe4efe495","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:08:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"282d37a4d28ecc22e444e81ad281f5302862774a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803bb4e2cc0a4fd59769e016fbe7ce2b : SUCCESS in 2m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1279004c58f34a8caa42be25470a43e9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5fed21fdfa754ca34d36ef5b0247394146bb961","revisions":{"d5fed21fdfa754ca34d36ef5b0247394146bb961":{"kind":"REWORK","_number":1,"created":"2021-03-26 13:33:44.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/04/783204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/783204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/783204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/783204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/783204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/783204/1"}}},"commit":{"parents":[{"commit":"1aa9073008083c35fa5ac6e44cfcafbce27a60db","subject":"Pure: Add default value to pure_host_personality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aa9073008083c35fa5ac6e44cfcafbce27a60db"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-11-19 16:21:57.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-26 13:33:44.000000000","tz":0},"subject":"API validation: Add cinder_host type to support ipv6 in manage","message":"API validation: Add cinder_host type to support ipv6 in manage\n\nThis will fail due to overly restrictive regex validation:\n  cinder manage hostgroup@cloud#[dead:beef::cafe]:/cinder01 abcd\n\nThis is because validation for this call checks for a valid\nhostname but needs to check for any string that would be a valid\n\"host\" in Cinder, which is not the same thing.\n\nCloses-Bug: #1904892\nChange-Id: I1349e8d3eb422f9dcd533c54f922f7ab8133b753\n(cherry picked from commit 8687925ce7b0e8f707006394f762a8f0a2ee60f3)\n(cherry picked from commit 51b92e9fa64baa3d362dd704ef064d042ef9541e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5fed21fdfa754ca34d36ef5b0247394146bb961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5fed21fdfa754ca34d36ef5b0247394146bb961"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
