)]}'
{"id":"openstack%2Fcinder~783660","triplet_id":"openstack%2Fcinder~master~I6824ba4fbcb6fd8f57f8ff86ad7132446ac6c504","project":"openstack/cinder","branch":"master","topic":"lvm-not-found","hashtags":[],"change_id":"I6824ba4fbcb6fd8f57f8ff86ad7132446ac6c504","subject":"LVM: Fix delete volume error due to lvs failure","status":"MERGED","created":"2021-03-29 14:44:04.000000000","updated":"2021-04-01 18:06:08.000000000","submitted":"2021-04-01 17:58:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783660-lvm-not-found","meta_rev_id":"a84f3adf7a7a3915941deba3597468d7ee94ee09","_number":783660,"virtual_id_number":783660,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 17:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-31 19:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-04-01 11:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-04-01 11:56:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-31 14:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 14:50:00.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-03-29 15:09:52.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2021-03-29 17:05:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 17:37:59.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-29 17:49:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-29 19:49:17.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-29 20:36:17.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-29 20:43:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-29 23:05:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 00:00:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-30 00:22:29.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 01:58:10.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-30 02:17:07.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-03-30 05:05:59.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 08:56:09.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 15:38:13.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-03-30 23:21:27.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-31 01:01:53.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-31 02:49:26.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-31 06:46:33.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-31 14:31:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-31 16:54:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c5066a09a0a2fa075de229a9b8e0a24eedecbfc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-29 14:44:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"facb0715fab21a8a552913d8652696fffd60c8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 17:05:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4f021c0098f4c4c837e3435ad22ef6c : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1f9fc0b3ed4aa9ad8f7a12b78bbbb9 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0709481fe22d409f9b8e404ead1fabc9 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04d77ffb089f40e8bce4a8f61120a831 : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4317fc2476864bcda0d6039c055b0573 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83847e692ad549708aaf7ab28825799d : SUCCESS in 16m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/8a8ba5d9fd0d4e6dadc83bd57d387735 : SUCCESS in 1h 03m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa86adfc6d27462abefd434bbe4736d5 : SUCCESS in 1h 25m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26dff65385d64d9796926570582b38e7 : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e6f5d4a896614055a092beebcf796de7 : SUCCESS in 14m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95090fbac7504221a6933eb627c0afdc : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72b8a9e117bb4c4985cf4508b3bb08bf : SUCCESS in 8m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/903a7762939b4254a2bedb72adf924ca : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d15693e1409147bf89d9278072894508 : SUCCESS in 53m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c2e58d6678f423b8bc39cee6fcbe60b : FAILURE in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e8b9a6dfc3843668b89c0dc3a336f1e : SUCCESS in 1h 32m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d62b9671a154d5587e078faa561da5f : FAILURE in 1h 30m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecb90f3a6fae4df29f47496002e071c9 : SUCCESS in 1h 44m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/377b4d83d88c49cbb26ec245f5c8c2f5 : SUCCESS in 1h 27m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6cedc6eaaa384d379bea182cb62b41d2 : SUCCESS in 1h 20m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6be310b23b79416aaeaf6e0be6813b94 : SUCCESS in 1h 32m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/744b3f94a8504251bde45b4e8aec1784 : SUCCESS in 2h 15m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4880ff15fec04742a59b8b0535d90303 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7c08ed52a95e9743ba5d5005dba622ca63f24b10","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-29 17:37:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/783660/1/check/dsvm-tempest-cisco-zonemanager-job/20a63eb/ : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"66da52857d720ef96ac1978cef4fb9bd334282e0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-29 17:49:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/783660/1/upstream-check/cinder-cDOT-iSCSI/d8823ec/ : SUCCESS in 42m 07s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/783660/1/upstream-check/cinder-cDOT-NFS/22ff035/ : SUCCESS in 38m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/783660/1/upstream-check/cinder-cDOT-FCP/7f0a3fd/ : FAILURE in 6m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"de2656518a4cdcb7028aa9b3c7e7829745a7e427","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-29 19:49:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/783660/1/upstream-solidfire-check/cinder-solidfire-iSCSI/6eee53b/ : FAILURE in 2h 41m 50s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b43d30543b90762cd89df95bcb8de62a647e0a7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-29 20:36:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/60/783660/1/check/dsvm-tempest-zadara-driver/e352703 : SUCCESS in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5b1d2f950d9e6dafed95de810cab473a09baefb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-29 20:43:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/783660/1/check/iscsi-driver-nimble/cb232f1/cb232f1.html : SUCCESS in 1h 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3de14819dea46ed6ecd1099049e2db8fd08a1d5a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-29 23:05:09.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/783660/1/check/tempest-dsvm-full-focal-py3/c9bff71/ : FAILURE in 2h 37m 35s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/783660/1/check/devstack-plugin-ceph-dsvm-focal-py3/864b13c/ : FAILURE in 1h 16m 17s\n- devstack-plugin-ceph-dsvm-focal-py3-newcloud https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/783660/1/check/devstack-plugin-ceph-dsvm-focal-py3-newcloud/78dfc21/ : FAILURE in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d8a58a8a7a580bde3bf2ec89ce4b5c64ff251e37","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-30 00:00:16.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/783660/1/check/DellEMC_Unity_FC/f5478aa/DellEMC_Unity_FC/1020 : SUCCESS in 1h 30m 36s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/783660/1/check/DellEMC_Unity_iSCSI/c6fa64c/DellEMC_Unity_iSCSI/961 : SUCCESS in 1h 56m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"856174e2ee0d34fc9631c296ec3cd8bffca7a2f0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 00:22:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/783660/1/check/DellEMC_XtremIO_FC/a2f555c/DellEMC_XtremIO_FC/1175 : SUCCESS in 2h 07m 28s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/783660/1/check/DellEMC_XtremIO_iSCSI/ebf0e48/DellEMC_XtremIO_iSCSI/1193 : SUCCESS in 1h 48m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc324c22881e01a2d24a96a933955db60b42d067","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-30 01:58:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/60/783660/1/check/DellEMC_SC_FC/7f904e6/DellEMC_SC_FC/539 : SUCCESS in 1h 59m 23s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/60/783660/1/check/DellEMC_SC_iSCSI/1233733/DellEMC_SC_iSCSI/536 : SUCCESS in 2h 09m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"5e6e2756095a29c250b26f2cb0c97f0154dbc92d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-30 02:17:07.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/783660-1/Lenovo-iSCSI : SUCCESS in 48:51.24\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/783660-1/Lenovo-FC : SUCCESS in 43:27.21","accounts_in_message":[],"_revision_number":1},{"id":"2a9cdba1f05390da551672118e45cab47225a92a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 05:05:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/60/783660/1/check/DellEMC_PowerFlex/9e5b48b/DellEMC_PowerFlex/689 : FAILURE in 2h 53m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"52f635918030a628b98605e15b0ed286cc8a8c03","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-30 08:56:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-783660-1/ns5_nfs : SUCCESS in 5h 11m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-783660-1/ns5_iscsi : SUCCESS in 5h 9m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a93d04ca202d13eb8babf3413fea21ae0bbb32e3","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-30 15:38:13.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/783660-1/PVME-iSCSI : SUCCESS in 55:47.36\n* PVME-FC http://os-logs.tristero.net/dell-ci/783660-1/PVME-FC : SUCCESS in 56:32.00","accounts_in_message":[],"_revision_number":1},{"id":"35c6acfbed6dfc01ee50daf5f4811b3f1bb2f427","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-30 23:21:27.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/60/783660/1/check/DellEMC_VNX_iSCSI/e323663/DellEMC_VNX_iSCSI/672 : FAILURE in 3h 11m 03s\n- DellEMC_VNX_FC http://publiclogs.emc.com/60/783660/1/check/DellEMC_VNX_FC/321dcfc/DellEMC_VNX_FC/720 : SUCCESS in 1h 25m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3bffff758e570d241934070d05db8ff6691c1e74","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-31 01:01:53.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/783660-1/Seagate-iSCSI : SUCCESS in 44:39.50\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/783660-1/Seagate-FC : SUCCESS in 39:18.99","accounts_in_message":[],"_revision_number":1},{"id":"b767d3d0a57565a30fb1a9059a4f7e40a86af4ca","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-31 02:49:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/783660/1/ : RETRY_LIMIT in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"05c1ada8c6c3abebef1493ca68623614ab4e1543","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-31 06:46:33.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/783660-1/HPMSA-iSCSI : FAILURE in 1:02:27\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/783660-1/HPMSA-FC : SUCCESS in 34:04.98","accounts_in_message":[],"_revision_number":1},{"id":"8d1869670870fc10cee8ff15e2e1542e3c8c00aa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-31 14:31:36.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"015913a9e714f27bc5eaa3a6d0eba36e672d21c6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-31 16:54:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/783660/1/20210331231656/NEC-FC : SUCCESS in 2h 19m 42s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/783660/1/20210331231656/NEC-ISCSI : SUCCESS in 2h 37m 53s","accounts_in_message":[],"_revision_number":1},{"id":"406d8a520df31685b8af2a464ccc9434f825f243","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-31 19:08:12.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks like a nice solution to this problem, and extends the kinds of retries we can do.","accounts_in_message":[],"_revision_number":1},{"id":"1b12e554c79f65ac085e4af24fb5b752ab92ed8e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-01 11:56:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"613c52aa28b094b5a0cf1271c62a7de546447a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 12:00:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b3ea6fa37663ddb49a48b95f070f8dba198189c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:38:03.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03c6b6f7c1067da3b7dfdb899eb677abacdcde69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:58:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6e40f8d27c544d9a7f8ae4934565fac : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf9a488febe4c86b5714612095de40c : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6de669b6ca084b78b41ee7ccafa39fd8 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c5a838921424781814b7ce15262f4ec : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560803133f4140a7ac0a789fd72e7168 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3ae9cb89488446749b538bd4dcf78cf2 : SUCCESS in 1h 05m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4421ff15e19141f8bedbccf3afab1be0 : SUCCESS in 1h 14m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34d0f4d81f6545d98c262be0ef718e77 : SUCCESS in 7m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1f6a2a391c344ca9d7a3d335d5eff16 : SUCCESS in 1h 22m 24s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c68c9e41d0c4e159146a0854671376b : SUCCESS in 1h 22m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a45b01534204f3b9ca6462f24087e5d : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9bf09ea6d10f5e90df044e27cb371d6b84655626","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:58:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a84f3adf7a7a3915941deba3597468d7ee94ee09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:06:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c8f22806524d6c9d52ee69dfce494f : SUCCESS in 1m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dea4d1cef58748cb9a57c0020febd112 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247","revisions":{"0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247":{"kind":"REWORK","_number":1,"created":"2021-03-29 14:44:04.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/783660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/783660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/783660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/783660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/783660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/783660/1"}}},"commit":{"parents":[{"commit":"2a959e6645379842880373dd9aad4d5ff3b6fd02","subject":"Merge \"Fix automatic quota sync for temporary volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a959e6645379842880373dd9aad4d5ff3b6fd02"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-29 13:47:28.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-29 14:43:10.000000000","tz":120},"subject":"LVM: Fix delete volume error due to lvs failure","message":"LVM: Fix delete volume error due to lvs failure\n\nSometimes we get an unexpected failure on the lvs command where it exits\nwith code 139, which is not one of the 5 predefined ones in the code\nECMD_PROCESSED, ENO_SUCH_CMD, EINVALID_CMD_LINE, EINIT_FAILED, or\nECMD_FAILED.\n\nWe\u0027ve seen this happen mostly on CI jobs when deleting a volume, on the\ncheck to see if the volume is present (``_volume_not_present``) and\nmakes the whole operation unexpectedly fail.\n\nWhen looking at the logs we can find other instances of this same exit\ncode happening in other calls, but those are fortunately to be covered\nby retries for other unexpected errors such as bug #1335905 that seem to\nmake the call eventually succeed.\n\nThe stderr of the failure is polluted with debug and warning messages\nsuch as:\n\n  [1] /dev/sda1: stat failed: No such file or directory\n\n      This has been removed [2] from the LVM code indicating it\u0027s\n      somewhat troublesome, but doesn\u0027t explain how.\n\n  [3] Path /dev/sda1 no longer valid for device(8,1)\n\n  [4] WARNING: Scan ignoring device 8:0 with no paths.\n\nBut the real error is most likely:\n\n  [5]: Device open /dev/sda 8:0 failed errno 2\n\nOn failure we see that error twice, because the code retries it in LVM\ntrying to workaround some kind of unknown udev race [6].\n\nSince the LVM code indicates that a retry can help, we\u0027ll retry on error\n139 when calling ``get_lv_info``.\n\nTo narrow down the retry we\u0027ll only do it on error 139, so we modify the\nexisting ``retry`` decorator to accept the ``retry`` parameter (same as\nthe tenacity library) and create our own retry if the\nProcessExecutionError fails with a specific error.\n\nThis pattern seems better than blindly retrying all\nProcessExecutionError cases.\n\n[1]: https://github.com/lvmteam/lvm2/blob/17f5572bc972c932bcf62fc2bff3029268ae0109/lib/filters/filter-persistent.c#L132\n[2]: https://github.com/lvmteam/lvm2/commit/22c5467add88268554a7d163a26892166dbc3104\n[3]: https://github.com/lvmteam/lvm2/blob/b84a9927b78727efffbb257a61e9e95a648cdfab/lib/device/dev-cache.c#L1396-L1402\n[4]: https://github.com/lvmteam/lvm2/blob/b84a9927b78727efffbb257a61e9e95a648cdfab/lib/label/label.c#L798\n[5]: https://github.com/lvmteam/lvm2/blob/b84a9927b78727efffbb257a61e9e95a648cdfab/lib/label/label.c#L550\n[6]: https://github.com/lvmteam/lvm2/blob/b84a9927b78727efffbb257a61e9e95a648cdfab/lib/label/label.c#L562-L567\n\nCloses-Bug: #1901783\nChange-Id: I6824ba4fbcb6fd8f57f8ff86ad7132446ac6c504\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
