)]}'
{"id":"openstack%2Fcinder~783857","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I0fd89028684c1406f427f4c97b120ece5cf2780b","project":"openstack/cinder","branch":"stable/wallaby","topic":"unused-quota-classes-stable/wallaby","hashtags":[],"change_id":"I0fd89028684c1406f427f4c97b120ece5cf2780b","subject":"Fix automatic quota sync for temporary volumes","status":"MERGED","created":"2021-03-30 12:41:48.000000000","updated":"2021-04-04 03:35:13.000000000","submitted":"2021-03-31 07:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783857-unused-quota-classes-stable/wallaby","cherry_pick_of_change":781104,"cherry_pick_of_patch_set":2,"meta_rev_id":"da031ec6e0f8a2de1eb3abe2fb224c0f37986071","_number":783857,"virtual_id_number":783857,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 07:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-30 14:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-30 14:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-30 14:51:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 12:41:48.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-03-30 12:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 13:09:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-30 13:45:33.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 13:45:54.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-30 13:56:10.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 14:06:56.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-30 14:51:24.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-30 17:27:24.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-04 03:35:13.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4f6940a6cca3b7db9546aa9e76ab3721ca56f038","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-03-30 12:41:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a15f6818f52334688206548b9f28390759fa657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 12:58:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daffe45693e54328b9440bcfc1be3006 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6029e9602f94ec4ae1b0201e80c39d1 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b059e8998b42d889f66e9c1025b432 : SUCCESS in 10m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06757e8efb2a4e6eab8313fb0c35b507 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a626f8a2ce4e4ba29d6696b6912e92 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06c3be31f1a4fd1b477b3f9a0d808aa : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e27acd35df4537a745e19d1eeb8fe0 : SUCCESS in 7m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f504c3ed394e456f99d9d7d0f0598a55 : SUCCESS in 12m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98f3c4f772d44547b3788f5656dfd820 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c003988aa8544d5a9f2e3011c610a40e : SUCCESS in 7m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7295f96d85548e08e57d5a43b5144bf : SUCCESS in 7m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2ac1113546941688383c95459530698 : FAILURE in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8aeb9f5c0757c6787a34b80ec5fe34dc48dbab0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:05:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffa1d5019eb74f5ab71ced10f525fd83 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b72c7043c7204d8ca3f12a7735381e9a : SUCCESS in 18m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7db2766d1c5d5c209ba2523feede25be27f18eee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-30 13:09:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/57/783857/1/check/dsvm-tempest-cisco-zonemanager-job/017270a/ : POST_FAILURE in 21m 18s","accounts_in_message":[],"_revision_number":1},{"id":"39c8a7a4e09c46e73679467db8a1af3b1199abca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 13:45:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/57/783857/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/15c0532/15c0532.html : FAILURE in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fc545699523cb277dfa87e1a74b9a23c6b9cf60b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-30 13:45:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/57/783857/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f4ac106/ : FAILURE in 46m 07s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee137b711850a7c076bb88cc3cab788092ae359a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 13:46:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/57/783857/1/check/iscsi-driver-primera/fafd9ce/fafd9ce.html : FAILURE in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9c601f6fbb2ef8d6742bf0ec41bf7acda3a0710e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-30 13:56:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/57/783857/1/upstream-check/cinder-cDOT-iSCSI/8469a48/ : FAILURE in 3m 04s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/57/783857/1/upstream-check/cinder-cDOT-NFS/790ed78/ : SUCCESS in 38m 42s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/57/783857/1/upstream-check/cinder-cDOT-FCP/aa93ea1/ : SUCCESS in 56m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"612684948de2ef4868f7310b7668379e5f553889","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 14:00:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/57/783857/1/check/iscsi-driver-nimble/c97f678/c97f678.html : FAILURE in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1bf54b3c30e46dc36a930c018d874b4949541690","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-30 14:06:56.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good.  I think it is important to get these fixed.","accounts_in_message":[],"_revision_number":1},{"id":"d2787a69ef021dcb3628744c27cd6e068e6d5e2a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-30 14:51:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nWe\u0027ve agreed that this is appropriate for inclusion in the wallaby release.","accounts_in_message":[],"_revision_number":1},{"id":"1b47a055e8f488ae420790e404193c834cbb10dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 14:58:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5b4805a2f1d5ecbf68a158b2999d523c9494a45","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-30 17:27:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/57/783857/1/check/dsvm-tempest-zadara-driver/d1aff79 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9ac7cf3d2de9bb2dbf5272bfa650aee828cfc6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:21:05.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d268976b6d17f86085ca612e89a9a852596169db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:23:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c22691d1be9d48dfa21224fb992e2859 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0fd95ac8fb40a4b94537f3a3b3cef5 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44358b02d1c44ee0b4254745ed4c879a : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/790e54f6f616424f90cc6ff8cbe6f95e : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52dc512653244ccb30a81ef0901524d : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418127b90cbe4361bed5fc45bd2bbaff : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"91f44f4f94c9a9439004613239bba4706eb2dac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:23:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"206431c1565508ed7d535ed0117feb4b00678612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:37:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ca1441564c46f1b1a07f802458b5ad : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eaaa1104f3044cb899b8ef46e5d2b43 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"da031ec6e0f8a2de1eb3abe2fb224c0f37986071","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-04-04 03:35:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/57/783857/1/ : FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87a9338466406bc0575bc5fc00ae9cc2ebdb342e","revisions":{"87a9338466406bc0575bc5fc00ae9cc2ebdb342e":{"kind":"REWORK","_number":1,"created":"2021-03-30 12:41:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/57/783857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/783857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/783857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/783857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/783857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/783857/1"}}},"commit":{"parents":[{"commit":"9623f83e011a08f408e718225a0f24c061542b4b","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9623f83e011a08f408e718225a0f24c061542b4b"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-15 14:12:13.000000000","tz":60},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-03-30 12:41:48.000000000","tz":0},"subject":"Fix automatic quota sync for temporary volumes","message":"Fix automatic quota sync for temporary volumes\n\nWhen using the automatic quota refresh via `until_refresh` and `max_age`\nconfiguration options the calculated quota usage by the refresh will not\nbe correct if there are temporary volumes (such as those created from\nsnapshots for backups).\n\nNormal quota usage calculation does not work like this, as it has admin\nmetadata stating that those are temporary volumes and are not used to\nincrease/decrease quota usage.\n\nThis patch fixes this by modifying existing filter that ignores\nmigrating volumes so it can also ignore temporary volumes based on their\nadmin metadata.\n\nCloses-Bug: #1919161\nChange-Id: I0fd89028684c1406f427f4c97b120ece5cf2780b\n(cherry picked from commit 8f03b26f50dafd161d5de5a6964fc9803177c1ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87a9338466406bc0575bc5fc00ae9cc2ebdb342e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87a9338466406bc0575bc5fc00ae9cc2ebdb342e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
