)]}'
{"id":"openstack%2Fcinder~783865","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I2a521118598de336148a56c1d5352107ece80721","project":"openstack/cinder","branch":"stable/ussuri","topic":"bug/1915800","hashtags":[],"change_id":"I2a521118598de336148a56c1d5352107ece80721","subject":"Add ports filtering support to Dell EMC XtremIO driver","status":"MERGED","created":"2021-03-30 14:40:18.000000000","updated":"2021-05-03 22:00:10.000000000","submitted":"2021-05-03 21:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783865-bug/1915800","cherry_pick_of_change":775798,"cherry_pick_of_patch_set":2,"meta_rev_id":"0f8d540acf412f384ba49c475c8b18084bc29a35","_number":783865,"virtual_id_number":783865,"owner":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 21:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-05-03 20:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":1,"date":"2021-04-13 09:45:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-13 15:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-05-03 20:42:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 14:40:18.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2021-03-30 21:06:04.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 21:40:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-30 22:13:08.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-03-30 22:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 23:27:46.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-31 06:07:19.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-31 13:18:53.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-04-02 01:42:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-04 17:34:23.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-05 08:36:23.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-07 20:50:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-07 21:09:01.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-13 07:56:18.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-04-13 08:55:58.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-04-13 12:10:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-13 15:35:19.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-05-03 20:42:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"858407ef44a71f380a4dc6cf7b72f46720a8f25a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-03-30 14:40:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6e82f2bf264db032b1dc3b93b896dcb6c9140058","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-03-30 15:10:14.000000000","message":"Topic bug/1915800-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"6e6492c1f7c0c0455282139b98a91969b310229f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-03-30 15:10:17.000000000","message":"Topic set to bug/1915800","accounts_in_message":[],"_revision_number":1},{"id":"92f9adb48704e19ad72702a77422841c8e7a02c2","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-03-30 19:48:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5e183e6d20ec3e8caf7e4965ddbd153403b47217","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-30 21:06:04.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic/9f73bdd/ : SUCCESS in 2h 02m 36s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic-py3/000ad55/ : TIMED_OUT in 2h 45m 37s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/devstack-plugin-ceph-dsvm-bionic-py3/f7e4057/ : SUCCESS in 2h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ef57807022be2e72573cfffbfef3938be0fa5747","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 21:40:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/65/783865/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/41c2322/41c2322.html : FAILURE in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4b239bc6644f2535d0c91214beff871493245cc6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-30 22:09:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/65/783865/1/check/iscsi-driver-primera/7c29c5b/7c29c5b.html : FAILURE in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4aa537bab7915287741082b973795d281812fd70","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-30 22:13:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/1/check/dsvm-tempest-cisco-zonemanager-job/6128697/ : FAILURE in 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a311b42079de9c1b81fd146a59d6f8efed91d4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 22:22:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f327c9a504364c81b2aafeea226b44fd : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319855a1dc37466a92e7eda1dcb98ed7 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8faef260886a4542a801cd2b63699891 : SUCCESS in 11m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a3fde7f93f54768a37edd4f6e154bdc : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b6e1d1d0684bfcb27c10a513ad3806 : SUCCESS in 17m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/bcf5b8f55e61407b9b502521464f444e : SUCCESS in 1h 07m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2edf8e3afc2b4f7d8bfb08bf25b3a4dd : SUCCESS in 1h 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82b8e61f71aa46df8e2edb63b906e205 : SUCCESS in 8m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a72e7fdac7c84f799274d772f07a712e : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/205e772addd9444a813de286451cafda : SUCCESS in 7m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6f80354c3f934013b9812a0449abfe04 : SUCCESS in 8m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e7f05ed2c8644ceb6fcb0e14dc5619b : SUCCESS in 38m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fffc21ff426f46099621fd8a76dbeedc : SUCCESS in 7m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/274d96fe0cd74c059309bed792dbf63e : FAILURE in 1h 32m 14s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2518931c1c3144dfa348edc498ea4f82 : SUCCESS in 1h 34m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ceec57a474bd4a29a1fe0bb21d6cc066 : SUCCESS in 1h 19m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b96e6f68733c45f99cf5b75e6811403b : SUCCESS in 1h 24m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a72123c82f304814948fe5b09b531283 : FAILURE in 1h 08m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ec50cd3faf740f28314483f119b7e3c : SUCCESS in 2h 02m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/688d278e995c425aaa10e90174402382 : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"21c08ef00f24638e6211d38e56b28daeb46290a7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-30 23:27:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/783865/1/check/dsvm-tempest-zadara-driver/0d3c698 : SUCCESS in 1h 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"447c63a5696ef5fce3db411b2978071ae124fec3","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-03-31 03:37:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3e35f69edc4fe4c1d47e00c8181d7bb25003028","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-31 04:10:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/1/check/dsvm-tempest-cisco-zonemanager-job/cb10755/ : FAILURE in 27m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7b6785231d0017034cbb1bf5616bb6585c966672","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-31 05:26:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/65/783865/1/check/iscsi-driver-nimble/445315a/445315a.html : FAILURE in 1h 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"73716e76302949a16278bc934c8665f07b80ccdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 05:29:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66509c1e99734d1193b91bcd08186f2d : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f96e7504754704aac902cde6fba8d0 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6be161458e2a4c5a9d8d95ff831a8598 : SUCCESS in 10m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df0254254b904be787e9d9e34f432479 : SUCCESS in 13m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c652dc4ae842e5b8e3d1314aaa6a33 : SUCCESS in 16m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b4b4802a5a6a46e0a4c9319c35be5fdb : SUCCESS in 1h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88b9107db3a942d1be0995ec375f5f09 : SUCCESS in 1h 00m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9395c9866d043e28a77d857df66a338 : SUCCESS in 6m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f80aac0f79de404d8176148efe4f7ecf : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32dce6dee49f438da5dbf36df4b3a99b : SUCCESS in 6m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/957ac18995fd4362bb4ad63fbe4f47b1 : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0338775c07674e13bfd878b0859f1f94 : SUCCESS in 50m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7df3b056e1fe4a34941f5fa2130494cb : SUCCESS in 10m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/987c169bafa94c36b1428733d1cba4c7 : FAILURE in 1h 47m 09s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f27bfa690f1c4eca884ca6924aba45ee : FAILURE in 1h 48m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6fb47bc83c64bb0ade66c07e3559b65 : SUCCESS in 1h 11m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a00d5002ff384299ba1a9b471d9e7f6e : SUCCESS in 1h 17m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5df80b01111e4f1c8d2ba122f41b580b : POST_FAILURE in 1h 11m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e398488c136449b1859278b5da83e772 : SUCCESS in 1h 30m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/301f7ef7502849559beed794e3348623 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":1},{"id":"10e907c423ecebb1c0d03a5a70000473bf78e90e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-31 06:07:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-65-783865-1/ns5_nfs : SUCCESS in 5h 32m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-65-783865-1/ns5_iscsi : SUCCESS in 5h 22m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"15618c502e71f7e6e1cbf3715ecfad7799d9371a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-31 08:25:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic/b55b07a/ : SUCCESS in 2h 01m 38s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic-py3/644b2c7/ : SUCCESS in 1h 58m 16s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/devstack-plugin-ceph-dsvm-bionic-py3/8660f18/ : TIMED_OUT in 2h 37m 37s","accounts_in_message":[],"_revision_number":1},{"id":"13b010cfdb3095aae90c9c6b049ace36ca982a1f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-31 12:10:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/783865/1/check/dsvm-tempest-zadara-driver/1213c0c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66607224f6501cfdbd241264b4d5ee11633195b6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-31 13:18:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fcf40b1e80f4ebfbd0379d4fe8eaf7d9f83fef08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 15:51:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b037ba4748894232ac6520a2fa390868 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f04b51ffd949c29cac802ee47b722e : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7de88b5b6c874609ab5084a6c03337bf : SUCCESS in 10m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f85c2d58b1e14afeac7a9f61871f1df1 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d9cdda101941c1bf3853a6b5c693d9 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d1e90c54ae8e4a5abaa75ec16ede5773 : SUCCESS in 1h 05m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7e6369ee47f458289a21615e5b455ed : SUCCESS in 1h 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/475f1e8f6a8344b1a8a87cf67c03e90d : SUCCESS in 8m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3061d2a53d674c428afdb544734987de : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d09a85bc9bc543e28dae5a3f881db357 : SUCCESS in 6m 17s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a2868190c72c4737aa169d25c274e806 : SUCCESS in 6m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/448ddd2278ad41fc911fc99254e1fb06 : SUCCESS in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecfe4a80cb1749079f4c49c132f15572 : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65f36d1a1a9f4317b42af74509bfec6b : FAILURE in 1h 16m 58s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bdd06b92d6c435ab4d5b8fdae6b4a5e : FAILURE in 1h 19m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0209e1e611e44f0fa21aa1d5ccb504fc : SUCCESS in 1h 36m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40d0ebc11c1b4c33aadb1c1da8bedf29 : FAILURE in 1h 33m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a0a8d98efff345a5b5ab81fd9a6d3822 : FAILURE in 1h 13m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c502a204016f4421863a38a70b5daa05 : SUCCESS in 1h 57m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f9345253f0141c99b00501ec2195cb1 : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"674d3eb45c2a31017deda9d31d54a5062b852ba4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-31 16:04:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/1/check/dsvm-tempest-cisco-zonemanager-job/386a6c5/ : FAILURE in 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f50472476b2f29038b531a70111ee23af074910e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-31 19:25:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic/0146bd8/ : SUCCESS in 2h 08m 01s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic-py3/a12ced1/ : FAILURE in 1h 58m 27s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/devstack-plugin-ceph-dsvm-bionic-py3/6b84d82/ : SUCCESS in 2h 38m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c259b956300c8ab542767b4abed940fda9fe3265","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-01 02:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/783865/1/check/dsvm-tempest-zadara-driver/eb2c94c : SUCCESS in 1h 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca7275e48769aa0cf16542ecd0219289ea75e542","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-02 01:42:16.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/783865/1/20210402084830/NEC-ISCSI : FAILURE in 1h 52m 52s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/783865/1/20210402084830/NEC-FC : FAILURE in 1h 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b0c2242f7518faa1d2279aa723f87fbd034cf0e9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-04-04 17:34:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/65/783865/1/ : RETRY_LIMIT in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2889df6e4bb04ae7e47c426f34688fb463932cfa","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-04-05 06:44:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cd04d6a1f47292b37aade444eaba58932555f67","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-05 07:59:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/783865/1/check/dsvm-tempest-zadara-driver/4f148ed : SUCCESS in 1h 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e02f4ca2d0247cc17805e715b692ec8f3a5c5639","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-05 08:15:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/1/check/dsvm-tempest-cisco-zonemanager-job/475fee6/ : FAILURE in 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1a1eeeb28f6fb9294c0992c5c517e554ea49b0ca","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-05 08:36:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783865/1/iscsi/ : SUCCESS in 57m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/783865/1/smb/ : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2b01f231eb123d79c4c934485ebb133b3aed48dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 08:43:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f13145ee78440d2b123cfc267fef00f : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ac8b2387296483a85135830d33a87a0 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a4b093481e4411384b6e71dbaba2ca9 : SUCCESS in 9m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ba37584b1c04ff79caf88f901c6f234 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937c223c62734628bba20c4a1f2d51f0 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/68fc5c48e2294ce4bfce2015bdbbc9c5 : SUCCESS in 56m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb02a75373a44f2ab627b017a14e5128 : SUCCESS in 1h 22m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/999bd9b7bc31485197f8d427f7c5f550 : SUCCESS in 6m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0424a58e600c4e22b393cd7acdda986b : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3d1e8199f0848e2a6c0b5d936a956da : SUCCESS in 7m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/358ee2bf32e74fb791cad63037f18818 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/720d4080ed3746db8d96af74ae4b16cc : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dcc5fa96733a46dab46ab1489cb3d000 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d06ba6e10e7f4d508bee935ab46bca6a : FAILURE in 1h 10m 35s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c51482f8389d49a2897472590f54dc62 : SUCCESS in 1h 28m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91b31905783f4f28a5e79e8f2242af24 : SUCCESS in 1h 19m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/10627e7d2ee34d6c9450bcb15074d89e : FAILURE in 1h 54m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f92d770865bd4e4d9daf1279ce28ea40 : FAILURE in 1h 22m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2b7dabc637a43b5bb3fb2b17eab4d5c : SUCCESS in 1h 34m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15142347e3c9443987c38c0de92332f5 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c260d5b8a77a2cc9e93dbbac42f2ef7106c7e064","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-05 09:29:21.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic/a9d66da/ : SUCCESS in 1h 58m 46s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic-py3/f65afd0/ : SUCCESS in 1h 59m 31s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/devstack-plugin-ceph-dsvm-bionic-py3/99cf676/ : TIMED_OUT in 2h 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"df42306105e253af206ab6c389a8079f9ea12d5e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-07 08:09:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"68581b60ea41c952366df80b4d4aba7e00a4eff2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 08:46:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/1/check/dsvm-tempest-cisco-zonemanager-job/29f1892/ : FAILURE in 29m 14s","accounts_in_message":[],"_revision_number":1},{"id":"63e218d237240b5dcbb5dc9f8978fae945a8fae2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 09:42:25.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783865/1/iscsi/ : FAILURE in 1h 32m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/783865/1/smb/ : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7ff76cef359311a50f20dc4771b38086db92a9c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:58:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57c656e3d9bc4171ba9572732172b78e : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad2b61290a342319c2ed2aa43d17207 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d3f9fc89fc347038cbbfde2b3154c2d : SUCCESS in 9m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6eaece9a5db049e496ebeaf1b85e5805 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa2d28ae1c84cc48add06366700542d : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a03d3a891fae4de9b7ba9a98ef3c02a0 : SUCCESS in 1h 18m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdc3ea1623284b66a071e472e48b6f16 : SUCCESS in 1h 17m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/274638e9798f499ca19bf3d6b13d58ac : SUCCESS in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/904a93156e9b4e5b8e765cdf8c7b6374 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8567b1c91fe4910947388009a2815ed : SUCCESS in 6m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/687c55680f6d460dbca68dd93c513e5f : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/06c14ce020f149948a96534bc9bfdb12 : SUCCESS in 37m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0fea48dfef204bf48a90fa23b6c8d686 : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5e38be0b82f94dc7aa79a8e543783536 : FAILURE in 1h 29m 12s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/195baccf374d41ecb6e8e4098786bd65 : SUCCESS in 1h 31m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3fe7dac518614c31aeaedbea2d1abc73 : SUCCESS in 1h 31m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3af4d36238b64523864447ec562b27e7 : SUCCESS in 1h 43m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f440498471c4209a884a0e5f6520313 : FAILURE in 1h 17m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/393558522fdc4d858b96ab577bea6fe2 : SUCCESS in 1h 46m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8868b8dc74c44e8baea2658c6932a5f8 : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":1},{"id":"11e60123954eb17abd410d58187c7591fdeb3c07","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-07 11:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/783865/1/check/dsvm-tempest-zadara-driver/94dc704 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a18e7d3196691c5ce4671c77d261860b2b4af4cc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-07 11:08:17.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/100af44e8a64454c9817ec5f2cbe4cd3 : DISK_FULL in 2h 37m 11s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/7bb6f314ed5a4305846a909db5592367 : DISK_FULL in 1h 21m 21s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/devstack-plugin-ceph-dsvm-bionic-py3/3635e36/ : TIMED_OUT in 2h 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"10b02072861ba2ad84539297ffea3b51d5707c7a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-07 14:30:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51cf50f75b1fae4a84bbb107d441a74d97c34e06","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 16:11:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783865/1/iscsi/ : SUCCESS in 55m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/783865/1/smb/ : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9bd7ade9fb9fee265c1e4179fd15c2d7cf634722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 16:43:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef5abbed9cac47a6b80527ced3612e4d : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8ed660461046e5b7ca606e4f382958 : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f610948dbe34ca7a95bc1f6fec14fad : SUCCESS in 10m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dc5c4a65adc440ebd0a930d9deccb4d : FAILURE in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe3c3cd6d31447e9eb831d8de9cbf12 : SUCCESS in 16m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ae4e461235304d7a9ba1d6e601a1012d : SUCCESS in 1h 11m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/658e90fb1f0345eabb5a4acfb1995eef : SUCCESS in 1h 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa25bf3a71e54bf4b25996a2970dfbc6 : SUCCESS in 6m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a073827cf804e9bb2cf55b41ea0429b : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b98066e6864434abcbe91c5059c7c90 : SUCCESS in 6m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/845313e5aee749fc8f70e9e78ec64556 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/316c7313bfe94559b7254dc996b55249 : SUCCESS in 43m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc5cad2e3746437d8c2e7d560034b108 : SUCCESS in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c67d8c3a733c4557b62e17b7a5208ef9 : FAILURE in 1h 35m 54s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/300dfc7c80754abaa1e2fb99b4cec07d : FAILURE in 2h 07m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c65f05743ec4c639ee8ba0dde374925 : SUCCESS in 1h 38m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b468fad0a9b643d1bea7a0c072af4f6c : SUCCESS in 1h 40m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/59c167ee8bc545619ef44fc5908d896e : FAILURE in 1h 08m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ee05ad0b7874435a91db4ba20a22428 : SUCCESS in 2h 02m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd6d9b7bf0147d6be1b0f7a36c73559 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c8ac1f55dd941c993f613a957d9f7807c19f4bb3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-07 17:00:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75e583994065bf7f8cee811e7dab1be34e0c4bba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 17:28:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/1/check/dsvm-tempest-cisco-zonemanager-job/43ed3ba/ : FAILURE in 27m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5ad36c6ddabf6044b12e58395e9d78ee90196483","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 18:06:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783865/1/iscsi/ : SUCCESS in 1h 05m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/783865/1/smb/ : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0a3c30a5da5481f7ae16d100c61d412c9cd417e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 19:33:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f921ac20fcc43269e2798ccb7690186 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4aa08e278b6433081c65ca44453b851 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef5a2f2cfd3d494c9f288924b235c375 : SUCCESS in 10m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84beb006e69543408e9e9756107aa733 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059790e7974b4d61adc83847939cc475 : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/2e186534ea6742ebaad7c99382a6ea8c : SUCCESS in 1h 10m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d2c01d670ee4167b6772e5abbf05f8f : SUCCESS in 1h 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a40c280db6de445591f3d33070315289 : SUCCESS in 8m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32f4cddf9371451cb1db25e17d8a8cf7 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88d3cb34d716494aa673b9c6095c736a : SUCCESS in 7m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cf34169c5c8e48c8b991dd44338b1d03 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ffdc977c007c4c3698e32533f725ac72 : SUCCESS in 43m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1d20f8830c5464f9635b4cf72865181 : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d6fc2e9b00d43d393bda7abbc2cea5e : SUCCESS in 1h 27m 05s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c03154624d74eeeaeb79e9432a08a7c : SUCCESS in 1h 26m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0b255e0609fb4436b1a55b16d51670e2 : SUCCESS in 1h 24m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38011b06baed4207a256464f293beadc : SUCCESS in 1h 17m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b04f806c8ec54109a202e460c19ffe36 : FAILURE in 1h 14m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44c2bb9057904de7953bcf6f4e28db5c : SUCCESS in 2h 05m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0382b33205154c5099bcdc96ed0794cc : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"322f125d35fee31990cf0e4c1cefb060647c4e1a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-07 19:44:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic/ccb830d/ : SUCCESS in 2h 03m 03s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/tempest-dsvm-full-bionic-py3/1c668f1/ : SUCCESS in 2h 39m 47s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/1/check/devstack-plugin-ceph-dsvm-bionic-py3/2e9f47b/ : SUCCESS in 1h 52m 23s","accounts_in_message":[],"_revision_number":1},{"id":"db26fbcfbc5fbe783ab26a83ade880e75ad1ce1e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-07 20:50:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/65/783865/1/upstream-check/cinder-cDOT-iSCSI/7a1d979/ : SUCCESS in 36m 08s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/65/783865/1/upstream-check/cinder-cDOT-NFS/96b698c/ : SUCCESS in 38m 36s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/65/783865/1/upstream-check/cinder-cDOT-FCP/8ebcc3a/ : SUCCESS in 1h 10m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e75d38b43fabf5b56b294fd60254bd40f02a2a9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-07 21:09:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/65/783865/1/upstream-solidfire-check/cinder-solidfire-iSCSI/376b08f/ : FAILURE in 1h 34m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"a4447e8318361170a5141152ade1d4bf996048c2","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-07 23:06:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/783865/1/check/dsvm-tempest-zadara-driver/2a2ea26 : SUCCESS in 1h 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e51bf25b169b59b9e440ff985a31aece52c26cc","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-04-13 07:56:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e30435066e4acf8f210726ae9d727ef0e536b1f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-13 08:55:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"44a4ecbc783a5d5a1aa14b4ae8917fa409baad83","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-13 08:58:16.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease follow the cherry-picking rules:\n\n\nhttps://docs.openstack.org/openstack-ansible/latest/contributor/contribute.html#backporting\n\nhttps://docs.openstack.org/project-team-guide/stable-branches.html#proposing-fixes\n\n\n(i.e. this one should be cherry-picked from the victoria fix)","accounts_in_message":[],"_revision_number":1},{"id":"59dc1e7cdc9a0e340f03f5c4ffe6a95c1cb9919a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-13 08:59:40.000000000","message":"Patch Set 1:\n\nIf you have your local clone and you have the stable/victoria patch applied (git review -d 778982), switch to stable/ussuri and cherry-pick -x that commit.","accounts_in_message":[],"_revision_number":1},{"id":"109f81b4bf71e09f47e7d1d31460bdc8bdf6f520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-04-13 09:13:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f75a1a8cd67ad10384f6e13798a0c3050afe30c5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-13 09:45:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1f77da5b57461e2d5d295bc66a1a530652a80dec","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-13 09:46:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/783865/2/check/dsvm-tempest-cisco-zonemanager-job/cf27094/ : FAILURE in 26m 27s","accounts_in_message":[],"_revision_number":2},{"id":"36d102ea6f6825377277c43367d5d45b75c2f03d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 10:20:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/65/783865/2/check/iscsi-driver-primera/5e74302/5e74302.html : FAILURE in 1h 04m 42s","accounts_in_message":[],"_revision_number":2},{"id":"88e3bbc5fb4657d18abf9f4c8bdbce9354ea7c61","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-13 10:21:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/783865/2/iscsi/ : SUCCESS in 1h 06m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/783865/2/smb/ : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1d4d616ef9034291a3858125a37c0ee6dba04a41","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 10:38:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/65/783865/2/check/iscsi-driver-nimble/0432a00/0432a00.html : FAILURE in 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"fbb60303688d57d3b107edac3819d686f24a3253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:28:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a90c6ce2644436a1561ac17193d42d : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befa93c6db374b99b2003da699b8c4b4 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8efec2219d5241148e6a46887f6e87b0 : SUCCESS in 8m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fbad80a57b54242b5392e59efdc6354 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553d13fe275c450290e0bd330a8662c7 : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/63b129ffca534a9199048be45ab15146 : SUCCESS in 1h 21m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3093a65d33345f88f95f9b92f358d1c : SUCCESS in 1h 25m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ffdf1056f064e378292484256a22a47 : SUCCESS in 7m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c81bbd3558454adb8394a20535e96b28 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37ff2da4d3a74c009a5e863e265e0968 : SUCCESS in 6m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f764f88899554ff2ad4f5f8178ca0142 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e374893195c467ab01c6c1ab459318b : SUCCESS in 36m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8157ac3f73b4f5ba4a57f407b1b1064 : SUCCESS in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/33f3285773404b9db35ae3036f187c84 : SUCCESS in 1h 16m 03s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f3538e38843d4706b036f7c4897e716d : SUCCESS in 1h 38m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98fc39f77bf24e9b89a9e8b5e8d84afc : SUCCESS in 1h 19m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f5ba8055745e48a092ee79e1ffdbbd2b : SUCCESS in 1h 33m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b10e5d710705498c88b2d56650b89aa1 : FAILURE in 1h 08m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f877c65ab6ea413d9b7d3295b1597031 : SUCCESS in 2h 09m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/693c38bd57c84bef9147c29d3b5c5db0 : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a709dcf0d1f4d3b26aa9319d65ce75ea2f8f08e6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-13 12:10:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/65/783865/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/290dc0d/ : FAILURE in 13m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/65/783865/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cdfaa31/ : FAILURE in 13m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e3dd4a2b6990e6c1615ec3f0f51a0dce52942c4a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-13 12:31:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/65/783865/2/upstream-check/cinder-cDOT-iSCSI/41ee6c5/ : SUCCESS in 36m 56s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/65/783865/2/upstream-check/cinder-cDOT-NFS/bb47bd2/ : SUCCESS in 39m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/65/783865/2/upstream-check/cinder-cDOT-FCP/0afe811/ : SUCCESS in 1h 02m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5895a1ac70ac7cc4ce44e13ab0b38eedeef742d2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-13 13:20:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/65/783865/2/upstream-solidfire-check/cinder-solidfire-iSCSI/440df34/ : FAILURE in 1h 51m 31s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"a9b503ed646bfe403d3267012354a67a0824ee3e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-13 13:47:15.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-65-783865-2/ns5_nfs : SUCCESS in 4h 32m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-65-783865-2/ns5_iscsi : SUCCESS in 4h 29m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"8f455f9dfef07d9f1cfb6a9bb15d7a752952e951","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-13 15:35:19.000000000","message":"Patch Set 2: Code-Review+2\n\nProperly backported.  Should be good.","accounts_in_message":[],"_revision_number":2},{"id":"6237ad241a666d530c7cc48211841ead02f0d044","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-13 17:02:27.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/783865/2/20210414003219/NEC-ISCSI : FAILURE in 1h 16m 07s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/783865/2/20210414003219/NEC-FC : FAILURE in 1h 30m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2cb8330d078846415e6a0aaf7de3d8b33f281914","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-13 18:33:26.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/edef02a188a04f1fb27ec9815125c81b : POST_FAILURE in 2h 10m 40s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/2/check/tempest-dsvm-full-bionic-py3/0d00c29/ : SUCCESS in 2h 10m 40s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/65/783865/2/check/devstack-plugin-ceph-dsvm-bionic-py3/4e7fca5/ : SUCCESS in 2h 16m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a9c3afe0be480235ca0e717f83afbb9e711ae47f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-03 20:42:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI\u0027ll repeat my comment from the stable/victoria backport:\n\nI agree with Jay that this is OK to backport.  This only affects one driver, and if we don\u0027t backport it, that driver is subject to the timeout issue when there are targets in the array that aren\u0027t connected to OpenStack, which I imagine will be the case for a really large array that\u0027s being used for multiple applications.  Further, the new option defaults to what we have now, so it won\u0027t affect current users who aren\u0027t seeing any issues.  It\u0027s low risk and high reward, so a good backport.","accounts_in_message":[],"_revision_number":2},{"id":"5b0ca84af6642c3b5eabb99ebf1411715139ec53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 20:42:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"592b3f62d5ea6f9eb3bfeda4abc5ed6676a1290e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 21:58:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb958c2364c94f9596c8be5c1da65e53 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62eb9bdb114f43e893fba9ae47faec99 : SUCCESS in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d88f8dbeb93c4c15a83d09f96556fa15 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37a1cf2228f48488beadaa176291f50 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/69d34d4844334d13bfd3523c16fc9f8b : SUCCESS in 51m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c603139204784f9082d3a8ff454436fc : SUCCESS in 54m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27901e6bbcd14f0782531dc31b2f69d2 : SUCCESS in 6m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/747a7b882a054fda992f25407f72307e : SUCCESS in 1h 00m 01s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/252edce5ea5741f68be002c5b2dad696 : SUCCESS in 1h 03m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3c674dfc1684d0198a43063856711f5 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bce61343d4b9d17e09d7e5754320b3c940ec3bae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 21:58:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0f8d540acf412f384ba49c475c8b18084bc29a35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 22:00:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f74a57a69145aab31cb667721dc5cd : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aad83a6b77354b26afece603607ec4bf : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9c029c174ab872c19f3fe2da504d87015ff60a5","revisions":{"cb2389bf2145ba7a5db613baf3209cc82579b490":{"kind":"REWORK","_number":1,"created":"2021-03-30 14:40:18.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/65/783865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/783865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/783865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/783865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/783865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/783865/1"}}},"commit":{"parents":[{"commit":"1aa9073008083c35fa5ac6e44cfcafbce27a60db","subject":"Pure: Add default value to pure_host_personality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aa9073008083c35fa5ac6e44cfcafbce27a60db"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2021-02-16 10:29:24.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"vladislav.belogrudov@dell.com","date":"2021-03-30 14:40:18.000000000","tz":0},"subject":"Add ports filtering support to Dell EMC XtremIO driver","message":"Add ports filtering support to Dell EMC XtremIO driver\n\nIf there are some iSCSI or FC targets (ESXi for example) that\nare not connected to the OpenStack host,\nattach volume operation waits until timeout.\nThe XtremIO Cinder driver needs a new option to support ports\nfiltering.\n\nChange-Id: I2a521118598de336148a56c1d5352107ece80721\nCloses-Bug: #1915800\n(cherry picked from commit e175d64cac735e8845165d0e8b48c4a8f2402ea1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb2389bf2145ba7a5db613baf3209cc82579b490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb2389bf2145ba7a5db613baf3209cc82579b490"}]},"branch":"refs/heads/stable/ussuri"},"d9c029c174ab872c19f3fe2da504d87015ff60a5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-13 09:13:46.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/65/783865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/783865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/783865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/783865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/783865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/783865/2"}}},"commit":{"parents":[{"commit":"1aa9073008083c35fa5ac6e44cfcafbce27a60db","subject":"Pure: Add default value to pure_host_personality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aa9073008083c35fa5ac6e44cfcafbce27a60db"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2021-02-16 10:29:24.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2021-04-13 09:13:46.000000000","tz":0},"subject":"Add ports filtering support to Dell EMC XtremIO driver","message":"Add ports filtering support to Dell EMC XtremIO driver\n\nIf there are some iSCSI or FC targets (ESXi for example) that\nare not connected to the OpenStack host,\nattach volume operation waits until timeout.\nThe XtremIO Cinder driver needs a new option to support ports\nfiltering.\n\nChange-Id: I2a521118598de336148a56c1d5352107ece80721\nCloses-Bug: #1915800\n(cherry picked from commit e175d64cac735e8845165d0e8b48c4a8f2402ea1)\n(cherry picked from commit a357ca2755d613ba31556d2f42c120b4cafdc2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9c029c174ab872c19f3fe2da504d87015ff60a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9c029c174ab872c19f3fe2da504d87015ff60a5"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
