)]}'
{"id":"openstack%2Fcinder~784453","triplet_id":"openstack%2Fcinder~master~I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e","project":"openstack/cinder","branch":"master","topic":"mypy","hashtags":[],"change_id":"I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e","subject":"mypy: continued manager, scheduler, rpcapi","status":"MERGED","created":"2021-04-01 17:56:30.000000000","updated":"2021-08-21 10:32:40.000000000","submitted":"2021-08-20 17:17:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"784453-mypy","meta_rev_id":"6ea30486a85df569cc5e875853a57b74176caea5","_number":784453,"virtual_id_number":784453,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 17:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"all":[{"value":1,"date":"2021-08-20 15:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},{"value":2,"date":"2021-08-18 16:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2021-08-19 17:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-20 14:25:13.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-20 14:25:55.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 18:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 20:26:24.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-01 21:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 21:14:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 22:06:11.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-01 22:19:23.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 22:43:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 23:15:42.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-01 23:32:31.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-02 01:04:50.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 02:07:47.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-02 05:25:22.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-02 10:05:05.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 10:11:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 13:33:35.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 14:01:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-02 20:29:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 13:41:21.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 14:02:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-05 22:45:50.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-27 23:15:20.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 23:51:11.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-30 13:27:51.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-30 13:39:20.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 17:57:38.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-03 20:31:29.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-05-05 10:22:12.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-05-05 13:49:02.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-07-01 03:20:40.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 19:49:35.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-07-28 16:00:42.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-08-10 13:27:51.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-11 07:20:00.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-08-12 18:24:35.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-08-19 17:13:43.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"REVIEWER"},{"updated":"2021-08-20 15:06:53.000000000","updated_by":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"reviewer":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"state":"REVIEWER"}],"messages":[{"id":"5260f95140a32ec00ce62aec6d1328a2f6280357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-01 17:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a2690cfeb5cd3e94a2a7b239b0aa115f5ba89b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:26:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/725fe229cf0f46d88a70e41977661bb6 : FAILURE in 11m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b0f697815d248339a96c9ea424fd259 : FAILURE in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea124daa8b706edd4648e35eb16e1c6941872869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-01 18:59:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f474545e256999bc0df50caa99739f918ef7229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 19:27:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff173aa6fccd4f498e41fc791ff51831 : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/825c869839914bea9f1a64975363a47e : SUCCESS in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a04929aee540cee6cc374332c66f5d7b92cd6bd9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-01 20:26:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/2/check/dsvm-tempest-cisco-zonemanager-job/98514b9/ : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5ecb8b36a08ca0a6bc475d1b353b5fc63abab797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 21:08:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/532233ce32fa4c0e9c4cc140b97af6ae : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ba4d06efe44a8aaec83308038e9653 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11318a97d8c4428dbc452d453ee75c67 : SUCCESS in 16m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd2808fa7ad4cf58e546544a8fdeefe : SUCCESS in 16m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6c1758e63349f683617e4ded3facb7 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d30617eb164ec28d548462cf4156c1 : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/74d8c8489d98402893a553dd84be95c6 : SUCCESS in 49m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87e4d8650b66489d8b7cf2a8a1b35690 : SUCCESS in 1h 08m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a43465f628064500b696beb4ced33bbb : SUCCESS in 12m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e66b7590f96436d903f98e02b2c7910 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b58919397bab40cc94652593f8cc4a33 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/acf824ddfeaf46348e0197db331062c7 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25b50178576b4b16b1752546beee4e60 : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/509b299b5b094016ad2221267b9d2b2c : FAILURE in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6b62edd59434a25b93c9c49d7a89edb : SUCCESS in 1h 13m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/672c9b9b11ea47398bc5763010e4726b : FAILURE in 1h 19m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2883d0bafe7471bba228a1f8880c4e3 : SUCCESS in 1h 31m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/74ec5ba48dc3486aae4181e5ea62b265 : SUCCESS in 1h 14m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/64ee0d0abdac452784c1dbc86645ab39 : FAILURE in 1h 20m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2aae1077f9cd4c3f96ad2f3364317b0e : SUCCESS in 1h 21m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d7271c4c091e4624bbcc31c25ce1b42d : SUCCESS in 2h 03m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ba460b54b7c4ceba49859741a94e61d : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":2},{"id":"983be88a830fc734d65be4353a24ffe19db372c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-01 21:14:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/92d9cc8/92d9cc8.html : FAILURE in 1h 02m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2d7df33d0cf17440aa5b8c471b16107e506cd2d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-01 21:31:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/2/check/iscsi-driver-primera/ee1f367/ee1f367.html : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a841239a65f5a75709a235ce18404e3e6661f4f5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-01 22:06:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/2/upstream-solidfire-check/cinder-solidfire-iSCSI/d0c796e/ : FAILURE in 55m 23s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"35205e596d4c2532c619c0595a34a06dab0c13dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-01 22:19:23.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"18e8b58d9847fc4b6ecaa0a834ca0c291b73f7dc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-01 22:43:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/2/upstream-check/cinder-cDOT-iSCSI/77169a3/ : SUCCESS in 42m 14s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/2/upstream-check/cinder-cDOT-NFS/56361c2/ : SUCCESS in 44m 44s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/2/upstream-check/cinder-cDOT-FCP/d9caab0/ : SUCCESS in 1h 33m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a4c14f707aff3607ba3e6917209e178f88e1a6f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-01 23:07:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/2/check/iscsi-driver-nimble/d4181c4/d4181c4.html : FAILURE in 1h 20m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8796c4c205be10707235cea766eb0de9181d5d42","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-01 23:15:42.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-2/Lenovo-iSCSI : SUCCESS in 47:47.56\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-2/Lenovo-FC : SUCCESS in 44:31.67","accounts_in_message":[],"_revision_number":2},{"id":"5e4d01da1463f689e6116e0a776c221d977fe540","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-01 23:32:31.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/53/784453/2/check/DellEMC_SC_FC/d1a6f96/DellEMC_SC_FC/565 : SUCCESS in 2h 10m 32s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/53/784453/2/check/DellEMC_SC_iSCSI/2653538/DellEMC_SC_iSCSI/562 : FAILURE in 2h 23m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"634aac70eaf362e6f063287d00c81ca89d972008","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-02 01:04:50.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-1/ns5_nfs : FAILURE in 1h 22m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-1/ns5_iscsi : FAILURE in 1h 23m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"897ca17d347a8233bf4bc55ed8d8c75639bee21c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-02 02:07:47.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-2/Seagate-iSCSI : SUCCESS in 44:27.63\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-2/Seagate-FC : SUCCESS in 41:02.59","accounts_in_message":[],"_revision_number":2},{"id":"5691ac6ac82ca58948c421f5d5fdabb2b7ab9ab3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-02 05:12:07.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-2/ns5_nfs : SUCCESS in 6h 2m 41s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-2/ns5_iscsi : SUCCESS in 5h 43m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"8440a15e29b03ec9a88df7bc98636e2cf652b27c","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-02 05:25:22.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-2/PVME-iSCSI : SUCCESS in 58:57.47\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-2/PVME-FC : SUCCESS in 56:51.11","accounts_in_message":[],"_revision_number":2},{"id":"00bf10fe3a8a76712d9d855e8b254ee407927f2f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-02 10:05:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/53/784453/2/check/DellEMC_PowerFlex/749208a/DellEMC_PowerFlex/715 : FAILURE in 2h 59m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"234b07fa5fe346d78ddedb9948b15b8ad2c5612a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-02 10:11:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/2/check/tempest-dsvm-full-focal-py3/929a0f0/ : SUCCESS in 2h 19m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/2/check/devstack-plugin-ceph-dsvm-focal-py3/9441bc6/ : SUCCESS in 1h 49m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8a19b9fda136d14da6436eaf86e20f42142d8526","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-02 13:33:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/2/check/DellEMC_XtremIO_FC/b4617fc/DellEMC_XtremIO_FC/1218 : SUCCESS in 2h 04m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/2/check/DellEMC_XtremIO_iSCSI/a8d54c7/DellEMC_XtremIO_iSCSI/1237 : SUCCESS in 1h 48m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"c11c7433cffeb54c25de17d4af9a0237258efd3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-02 14:01:53.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/2/check/DellEMC_Unity_FC/3a3dc29/DellEMC_Unity_FC/1066 : FAILURE in 1h 58m 48s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/2/check/DellEMC_Unity_iSCSI/cf62c8e/DellEMC_Unity_iSCSI/1004 : FAILURE in 2h 08m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"48b2384624fc109afb58081f21f6a558e892905f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-02 20:29:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/53/784453/2/check/DellEMC_VNX_iSCSI/e6f0c2f/DellEMC_VNX_iSCSI/696 : FAILURE in 1h 16m 50s\n- DellEMC_VNX_FC http://publiclogs.emc.com/53/784453/2/check/DellEMC_VNX_FC/e8f9739/DellEMC_VNX_FC/745 : SUCCESS in 1h 42m 27s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e71027c013e74e60c37116e71ff45861642bbfe6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-04-03 13:41:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/53/784453/2/ : FAILURE in 3h 15m 59s","accounts_in_message":[],"_revision_number":2},{"id":"135c97572d602aac2c4d5c127064076f5210e931","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 14:02:23.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/2/20210403203458/NEC-ISCSI : FAILURE in 53m 10s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/2/20210403203458/NEC-FC : SUCCESS in 2h 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"57adc44b328e02352a0e4f1495b5fc768546c672","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 16:51:24.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2d0d72af71c1c5ddbbc08fe76ab053d580203962","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-05 22:45:50.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-2/HPMSA-iSCSI : FAILURE in 59:30.28\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-2/HPMSA-FC : SUCCESS in 33:52.41","accounts_in_message":[],"_revision_number":2},{"id":"8b7ad54280d1bbd017075f46ffae1a5ec4870cfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-27 21:33:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f66ff1534ff4e75ba933342fae65c94aa433cee5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-27 21:36:36.000000000","message":"Topic mypy-apr27-530b removed","accounts_in_message":[],"_revision_number":3},{"id":"815f602a51f79fe51025a406381867662fb51f1d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-27 21:36:45.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":3},{"id":"17ec975172dd940546a850cca69df4d2fbbce9ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 22:08:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e7dc187f8b944889be3ba150a99f9f5 : FAILURE in 28m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2e7b10864464d14a00297454d1b5b9a : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf36bbc03b58d01122942a340e7ef54ca791e593","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-27 23:15:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/3/check/cinder-storpool-tempest/ff22319/ : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ed0ee91786092640c62043a0c46f642ea0961293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 23:33:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5ea2167bce4454dbabb70789a6763f1 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809512ae1e1c478783a8eeaf4d8580ad : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9139f7b46a4d4767a0fd342735fe0333 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b627e470df9a45c2bbc0adc569db284c : SUCCESS in 18m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9ec4ab51f224761861ebf8535c67b98 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e7df4625874eebbabe0e592b63517c : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/93b89a91b222497ca46499309e132436 : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fb767e4a3084cb0ac82bf7daf413885 : SUCCESS in 1h 23m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b49a6e30197f46b58d2a845a82e3d3e6 : SUCCESS in 12m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e071002b2c64cf5a96bdecb94f55a67 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd592aa00ef04f9897d5e6cfaeffe229 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd7f44c391b942259d76df836f04ce3c : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/07417873740049f49af514ec285fd97f : SUCCESS in 43m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c36d5d924e864e3599baae66f1813256 : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/acdc47db652a4782ba0a02f845b14397 : SUCCESS in 1h 05m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f4ea1d996a24662ba2a391848d4edd4 : POST_FAILURE in 18m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/694af404063d4ca6b64c48a45cea1b9b : FAILURE in 1h 14m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e0ee276055c64550815dce4e4d0cf92d : SUCCESS in 1h 14m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0365959052d4c61b406315e7bb70957 : SUCCESS in 1h 30m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fcc9ba128e8941a290982f19d0159f10 : SUCCESS in 1h 19m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3793590c625a4d3a902cf0f1dbe1c770 : SUCCESS in 1h 54m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c7b0081501e43f3a57ad6866adc3039 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f94509ab1abbd670b6e6d5a979724b00eeb693e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-27 23:50:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/3/check/iscsi-driver-nimble/36ec1ca/36ec1ca.html : FAILURE in 29m 40s","accounts_in_message":[],"_revision_number":3},{"id":"560282179467e579aef15279fd46fb65d360dc5e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-27 23:51:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6306c1e/ : SUCCESS in 2h 12m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a7b75eb/ : SUCCESS in 2h 01m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a369e3f60a9c496211a8bb86952addc1eac45643","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-28 00:06:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/3/check/dsvm-tempest-cisco-zonemanager-job/bee9016/ : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1753f0d4198f1542b2e958d7576e69b2d2dd84f7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-28 02:14:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/3/check/tempest-dsvm-full-focal-py3/e8a342e/ : SUCCESS in 1h 50m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/3/check/devstack-plugin-ceph-dsvm-focal-py3/b0e0573/ : SUCCESS in 1h 38m 59s","accounts_in_message":[],"_revision_number":3},{"id":"c327dca606e45874bf6934ae3a09e38f2478bb5b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 02:49:33.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/3/check/DellEMC_Unity_FC/da5e433/DellEMC_Unity_FC/62 : SUCCESS in 1h 26m 44s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/3/check/DellEMC_Unity_iSCSI/6aecd64/DellEMC_Unity_iSCSI/61 : SUCCESS in 1h 42m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"66fcf12bd8f159ca8b330cea3572f41d2fe12d29","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-28 03:50:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-3/ns5_nfs : SUCCESS in 4h 46m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-3/ns5_iscsi : SUCCESS in 4h 43m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"80f81d487b439565833db5a071c106c6922b7611","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 03:55:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/3/check/DellEMC_XtremIO_FC/e24ba3f/DellEMC_XtremIO_FC/71 : SUCCESS in 2h 07m 06s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/3/check/DellEMC_XtremIO_iSCSI/3c11638/DellEMC_XtremIO_iSCSI/71 : FAILURE in 5m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"608ffb18773af81ba789dbae48ae745301d17279","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 06:57:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/3/20210428132401/NEC-ISCSI : SUCCESS in 2h 22m 45s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/3/20210428132401/NEC-FC : SUCCESS in 2h 33m 35s","accounts_in_message":[],"_revision_number":3},{"id":"4e8b1afc184c88cc603c41666f0477af05f75432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-30 12:26:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"174fd527a0c88233316018f4b10d3b7f325f9196","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 12:54:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0114b9ec71584fb9ae8fe74de54b2e85 : SUCCESS in 20m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f2a9c0f3def4dbd8ef4b3eb45579eee : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7addd9876eb1e9563dd498129076107e24637ff4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 13:24:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ef5a470/ef5a470.html : FAILURE in 54m 01s","accounts_in_message":[],"_revision_number":4},{"id":"523de17e230b89287ae1ac7dc016ffb803222df9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-30 13:27:51.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/4/iscsi/ : FAILURE in 59m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/4/smb/ : FAILURE in 57m 59s","accounts_in_message":[],"_revision_number":4},{"id":"7ccaf7a1c9db21e231df91278b926fd18b8a6bd9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 13:30:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/4/check/iscsi-driver-primera/fc5ac79/fc5ac79.html : FAILURE in 59m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6ab2ec80b41be2d8752ca5dbb1242919074523d3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-30 13:39:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/53/784453/4/check/dsvm-tempest-zadara-driver/a23cfa5 : SUCCESS in 1h 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d0f37ccc6bd9240066a4385f722b97ca557091cb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-30 13:45:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/4/check/dsvm-tempest-cisco-zonemanager-job/a555514/ : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b848a7263870c6532e241ee24e9f9f95e9147662","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-30 14:01:47.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/4/check/DellEMC_Unity_FC/4a4e51a/DellEMC_Unity_FC/78 : SUCCESS in 1h 20m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/4/check/DellEMC_Unity_iSCSI/6c0276d/DellEMC_Unity_iSCSI/74 : SUCCESS in 1h 32m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8f8ae2472546e50a7c7e48d52e7efe43390924a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 14:19:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/4/check/iscsi-driver-nimble/69f1294/69f1294.html : SUCCESS in 1h 49m 57s","accounts_in_message":[],"_revision_number":4},{"id":"15cf4ce71ec5d22beb1f9fc18cc9c96da08c4df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:35:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f345e59947f4841bfe9698184b975af : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06ab49852d34ce4985a25dc0d43f0cb : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f4b6b3d61a14c10be3d25a69694e132 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f795fbe0012e42ccace7a87b43919932 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f3b5515fc6d4e149eafc9603bb4bd94 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba66cf53dc054e18aae3c7bb2b4c0006 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c8b74f20cc474bbabeaffd0ecea2c5c6 : SUCCESS in 1h 30m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/537e42029a1f418e88eea978122b4eaa : SUCCESS in 1h 14m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/14939a1cc30c45298d877d8188d188cf : SUCCESS in 13m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e087915691aa43f798043f4d7a94cc34 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e62e43bcda7a4af59aad38950759d4f3 : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b554528d082425daf1f5588e3325c65 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/843d447489d54dbca67d9bfbb349abea : SUCCESS in 36m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cef4bf748a9941839af3a6a2bfc96d30 : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7092003525c54aca8abf8c7643b7fc7c : SUCCESS in 1h 11m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c7c62db52f554ce68f907e09f7456058 : TIMED_OUT in 2h 04m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7aa59d7e1e3b44f5a13fd54637a1a619 : FAILURE in 1h 25m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2053474a3aff4fd79fd2b37201c03785 : SUCCESS in 1h 06m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/252d2d7530c3498680705dd448de2efd : SUCCESS in 1h 09m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ee7ff4e1d55e459e9d62d8d299472184 : SUCCESS in 1h 12m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/554127df15bd4f58bf7f770b05f70e31 : SUCCESS in 1h 57m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af64ca2ab01142c6b19aa0b2a0648dde : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":4},{"id":"bd7883b4c2b57fd85b270f3b3714e0de2b85ba13","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 14:36:07.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/4/check/DellEMC_XtremIO_FC/6bcb9c0/DellEMC_XtremIO_FC/93 : SUCCESS in 2h 08m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/4/check/DellEMC_XtremIO_iSCSI/160d8ab/DellEMC_XtremIO_iSCSI/93 : SUCCESS in 1h 52m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"45b07bdeda8604a1f41a1d2297b9f08596cdbe10","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-30 14:42:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2654f80/ : SUCCESS in 2h 13m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/eb55f48/ : SUCCESS in 2h 05m 24s","accounts_in_message":[],"_revision_number":4},{"id":"99c4187efcf0654d0f80f4cb86fed6346ada0dbe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-30 14:45:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/4/check/cinder-storpool-tempest/8299d6a/ : POST_FAILURE in 2h 15m 28s","accounts_in_message":[],"_revision_number":4},{"id":"42f352d552091139a904ccae0ef0dee25e492589","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-30 14:56:26.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5a00482846e5444980da9a04b39af64a : POST_FAILURE in 1h 52m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f88c8722744b433185d48bf72e164ae1 : POST_FAILURE in 1h 53m 27s","accounts_in_message":[],"_revision_number":4},{"id":"9e16038a4916611e9779adf38b504626048fbac4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-30 17:08:34.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-4/ns5_nfs : SUCCESS in 4h 40m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-4/ns5_iscsi : SUCCESS in 4h 29m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf24a6787db6a8cba08471aa0b3d2633c6b734d2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 17:57:38.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/53/784453/4/check/DellEMC_PowerMAX_FC/4d26b12/DellEMC_PowerMAX_FC/39 : SUCCESS in 5h 28m 47s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/53/784453/4/check/DellEMC_PowerMAX_iSCSI/077a05f/DellEMC_PowerMAX_iSCSI/33 : SUCCESS in 5h 16m 19s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"97fbf58926ea69460516b1374aeffbc3cd850f15","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 21:59:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/4/20210501024710/NEC-FC : SUCCESS in 2h 22m 24s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/4/20210501024710/NEC-ISCSI : SUCCESS in 4h 12m 38s","accounts_in_message":[],"_revision_number":4},{"id":"e388e4fcb2767506721ec9d06915cddb7dc4d6ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-03 20:15:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a42aeba84c270dfc7a12cda3e446bd0b580637da","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-03 20:36:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2476b01/ : FAILURE in 18m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1856766/ : FAILURE in 17m 52s","accounts_in_message":[],"_revision_number":5},{"id":"9db573a8d9f21251534a1c414203d71d3056e294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 20:38:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f751493139874cc5a3d226d5de88a7b0 : SUCCESS in 13m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a24874087dff4cabbdf1aa74523053a6 : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58fd7c14083b8b6ea189c40d7eec61eaa21b338d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-03 21:27:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/53/784453/5/check/dsvm-tempest-zadara-driver/035e9c4 : SUCCESS in 1h 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f97a12c8f69faf3a66f6aa5f32ca51c301483e0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-03 21:31:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/5/check/dsvm-tempest-cisco-zonemanager-job/a182e03/ : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":5},{"id":"87eb1bc69704b7ef856ecb93b8361dfe21db93bc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-03 21:37:47.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/5/iscsi/ : FAILURE in 1h 21m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/5/smb/ : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":5},{"id":"a120e052502909bd32bc95492ac305474f50f23d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-03 21:38:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/5/check/cinder-storpool-tempest/4bde7ad/ : POST_FAILURE in 1h 20m 41s","accounts_in_message":[],"_revision_number":5},{"id":"4bf5319bae7b1a248cf12b293ceefb7b16533dde","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-03 21:46:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/5/check/iscsi-driver-primera/5f564bc/5f564bc.html : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":5},{"id":"9af56e01c2d776a937721850e50fd29e92f4ba43","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-03 21:49:54.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/5/check/tempest-dsvm-full-focal-py3/2610a48/ : SUCCESS in 1h 29m 53s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/5/check/devstack-plugin-ceph-dsvm-focal-py3/06a45b8/ : SUCCESS in 1h 25m 53s","accounts_in_message":[],"_revision_number":5},{"id":"70d179d8d3ff38441e1658b8c506897f1489c132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 22:10:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dddfa97096b4a2e8a1ebf59c13ed8c5 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868bd30828e641af8f651473922cd993 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0ee1baa9b4a46a2adb8c23d8dd7846e : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/212221eab0ec46cd8e9e25ee45eb2615 : TIMED_OUT in 41m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9524cac19b4d4b469d54d692175a1695 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cf1235c2c843f8a6dbfe379249eb3a : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/78fc0197745645d89728fff321e68eff : SUCCESS in 1h 11m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90c371b009404b53bdbb92d49c28f513 : SUCCESS in 1h 01m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1db0b6b2631421a8f22624282edd595 : SUCCESS in 14m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a083bfce5b5413fba1a12341c1600a9 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b23e02f2ddc450daf5d613fd53033cf : SUCCESS in 7m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0995f27ceb634ee89b387b10942f2871 : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e401f1d532714397b62c8225ee4005d5 : SUCCESS in 41m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/976b299bd02249be99d9403abb2478bb : FAILURE in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48f685c1c497410eb2c95a6c72285941 : SUCCESS in 1h 40m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5e4a48a6095a4b2fb2a9575ddfae86e6 : FAILURE in 1h 21m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b73df3de047b4a9a9c88ccb262c3efe1 : SUCCESS in 1h 36m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c691ab52e7d422c8ad9730345f7968a : SUCCESS in 1h 13m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/96516bb68f4042879f09e8420be0185e : SUCCESS in 1h 28m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68592e0a3b814113acab9ce4e3e88abb : SUCCESS in 1h 21m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5034b45d792e4a589fc58e777595087f : SUCCESS in 1h 53m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96291901b0f64ea99a7a907fbce02b0c : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a38c83623f258d4293a8ef2aa22652a5fb77000a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-03 22:25:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/da70963/da70963.html : SUCCESS in 2h 07m 35s","accounts_in_message":[],"_revision_number":5},{"id":"930bc5b5d7c3f8dfe69073f1abc94210379dde33","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-03 22:29:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/5/check/iscsi-driver-nimble/4a02a0e/4a02a0e.html : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":5},{"id":"ce7138669a574e419e139255dec3d77cc9e36a18","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-03 23:17:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_Unity_FC/c5b5feb/DellEMC_Unity_FC/102 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_Unity_iSCSI/cc7a2c3/DellEMC_Unity_iSCSI/96 : SUCCESS in 1h 25m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"acd12aa5ca172f86a17028fc61f4ebb00a57cc31","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-03 23:31:48.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_XtremIO_FC/147dde8/DellEMC_XtremIO_FC/114 : FAILURE in 2h 07m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_XtremIO_iSCSI/2a8898b/DellEMC_XtremIO_iSCSI/114 : SUCCESS in 1h 53m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"f915a04e330a70ec7693335ecbb9fc6a423fff8e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-04 00:15:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/5/20210504065129/NEC-ISCSI : SUCCESS in 2h 13m 43s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/5/20210504065129/NEC-FC : SUCCESS in 2h 24m 11s","accounts_in_message":[],"_revision_number":5},{"id":"d92212c25664fb1a54e656db16081b686982811b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-04 00:51:03.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-5/ns5_nfs : SUCCESS in 4h 34m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-5/ns5_iscsi : SUCCESS in 4h 28m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"c4e8fe0bc944c73272dbe74ef6ab7b38cb554560","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 07:05:39.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_PowerMAX_FC/beb3a81/DellEMC_PowerMAX_FC/57 : SUCCESS in 5h 41m 07s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_PowerMAX_iSCSI/f546844/DellEMC_PowerMAX_iSCSI/51 : FAILURE in 6h 55m 40s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"213c71352fbfc32b7208c0af05365e2cd4d7b619","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-05-05 10:22:12.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ca746bf27dbe8595dc74039ec111f0d5fd26b247","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-05 11:02:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/5/check/cinder-storpool-tempest/77aa726/ : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":5},{"id":"dc5d6b887d35e13862bc1dee091eeffddc4bf790","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 11:14:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de049d4c0a5d43fabdb2f66fb2f4e82f : SUCCESS in 42m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/250ae390c66a42f7a993e18e1d23dd36 : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cb0f17c7a6e3bc3ea7654802c55dbff967d49dac","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-05 11:33:11.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/5/iscsi/ : FAILURE in 1h 10m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/5/smb/ : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":5},{"id":"3020e62c0ecaf270dd192a9c72b53e48bbafe68e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-05 11:50:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8714124/ : POST_FAILURE in 1h 26m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8aa8ff8/ : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":5},{"id":"065b125802cb0fa4aac0df76f978804d683454a5","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-05 12:18:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/53/784453/5/check/dsvm-tempest-zadara-driver/8265dd1 : SUCCESS in 1h 09m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d6be084c6b5538fb9c3ce1a6994ae3c141793311","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-05 12:30:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/5/check/dsvm-tempest-cisco-zonemanager-job/e3bf5a2/ : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":5},{"id":"7b2a6e2cb87079b2237121c51096f48a621d916a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:01:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cb198cec1194600858d23ecafebc287 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d348432691d3466a917e6fcba0af2abb : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f40a2bf7f7e04b9c874c81fe69b9dd6b : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70bcf380573d49b5817b8f88cec545f3 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4b1a95f582f4573af001ebcaff8862c : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57df41f92c04c83b1a1226ed6fccc7a : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/96116cad1b0340f48b59e42e6f21d092 : SUCCESS in 1h 00m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afa5a9712641478e900d6a97c088fe7f : SUCCESS in 1h 06m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7fa3e85bb604ae098ced5f9f6819aa9 : SUCCESS in 13m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98e4bcd6b5d44f889173d793d5314f4e : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/863c644184d04bcbb05f3d1be980ce57 : SUCCESS in 6m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27d71a54173447f190e6aea434b79584 : SUCCESS in 9m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e015225f9a346f1b95050f05f5a534f : SUCCESS in 44m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6221ca9856a04aacbea53ab4b6252f4f : FAILURE in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aff6269965a64b08926264e0b75dda73 : SUCCESS in 55m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aadbd5d682074eadafaba1085f3d6411 : FAILURE in 1h 06m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc58791b37794c40a8d31c885e19d337 : SUCCESS in 1h 32m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/572aca2fff14449aaf293d3816649778 : SUCCESS in 1h 18m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38fdafda27854e85b65b16ce57009fb6 : SUCCESS in 1h 20m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7cfdf687d33344359734831449a29ac9 : SUCCESS in 1h 23m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5de66e63094441978d65e6a4322f576a : SUCCESS in 2h 17m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/259de29fbf374db9abeb86627b2d95f0 : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":5},{"id":"387406a419e8a793e9af31db5a65a92cb6194cda","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-05 13:49:02.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_PowerStore_iSCSI/f7987c2/DellEMC_PowerStore_iSCSI/47 : SUCCESS in 47m 09s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_PowerStore_FC/99e0366/DellEMC_PowerStore_FC/48 : SUCCESS in 47m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"22ca3dc07d3d511d599902157e60c234260f5d87","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-05 15:04:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_Unity_FC/0a50198/DellEMC_Unity_FC/118 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_Unity_iSCSI/d5e230b/DellEMC_Unity_iSCSI/109 : SUCCESS in 1h 37m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"277a7c05b9ff0467fb908dda1b7e3af003f85ee9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-05 15:18:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/5/upstream-solidfire-check/cinder-solidfire-iSCSI/cad739c/ : FAILURE in 2h 16m 16s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"28f979be5cb50186bffb9d24091aa6477b22191c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-05 15:26:23.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/5/check/tempest-dsvm-full-focal-py3/d5d1555/ : SUCCESS in 1h 39m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/5/check/devstack-plugin-ceph-dsvm-focal-py3/54882eb/ : SUCCESS in 1h 53m 23s","accounts_in_message":[],"_revision_number":5},{"id":"6464cc7ac2f559d106b8724b90114a584eacc3d4","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-05 15:33:00.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-5/Seagate-iSCSI : SUCCESS in 44:14.29\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-5/Seagate-FC : SUCCESS in 37:34.01","accounts_in_message":[],"_revision_number":5},{"id":"a640de6fc1483cfa0e775eb4024c4423c8f78c13","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-05 15:52:08.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-5/Lenovo-iSCSI : SUCCESS in 48:41.69\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-5/Lenovo-FC : SUCCESS in 43:20.53","accounts_in_message":[],"_revision_number":5},{"id":"a2416f7ee286945ca537267667bf81648923dc2a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-05 15:52:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/5/upstream-check/cinder-cDOT-iSCSI/ee3b6fb/ : FAILURE in 1h 15m 39s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/5/upstream-check/cinder-cDOT-NFS/f3f2f0b/ : SUCCESS in 2h 03m 13s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/5/upstream-check/cinder-cDOT-FCP/5ae248f/ : FAILURE in 2h 50m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"10987c88732928a5d38d4abd525d475577b4a15c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-05 16:14:16.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_SC_FC/f45a75a/DellEMC_SC_FC/70 : SUCCESS in 1h 56m 40s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_SC_iSCSI/0b2ee26/DellEMC_SC_iSCSI/67 : SUCCESS in 2h 03m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"91369e49947514a65169d1d88d8f8d7ede15d105","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 16:16:04.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_XtremIO_FC/1eac218/DellEMC_XtremIO_FC/125 : SUCCESS in 2h 11m 12s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_XtremIO_iSCSI/d3c40b3/DellEMC_XtremIO_iSCSI/125 : SUCCESS in 2h 00m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"bdd052d46aa72a210425d2e72b6a3d8861fa6ee4","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-05-05 16:24:47.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-5/PVME-iSCSI : SUCCESS in 53:15.86\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-5/PVME-FC : SUCCESS in 53:57.83","accounts_in_message":[],"_revision_number":5},{"id":"ebc620c184f0b99f07130c396d8859d043891fca","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-05 16:49:23.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-5/HPMSA-iSCSI : FAILURE in 1:26:35\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-5/HPMSA-FC : SUCCESS in 33:31.73","accounts_in_message":[],"_revision_number":5},{"id":"eba685701f89b9350fedbe1a2d15a717dea698e8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 17:22:35.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/53/784453/5/check/DellEMC_PowerFlex/21c679b/DellEMC_PowerFlex/50 : FAILURE in 2h 44m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"7d32f4fc5bd22bb9ce8779e1c697d311377845df","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 18:19:29.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/53/784453/5/check/DellEMC_VNX_iSCSI/0e88087/DellEMC_VNX_iSCSI/74 : SUCCESS in 3h 01m 15s\n- DellEMC_VNX_FC http://publiclogs.emc.com/53/784453/5/check/DellEMC_VNX_FC/b656d1c/DellEMC_VNX_FC/77 : SUCCESS in 1h 25m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"35c8ddcf0a7973262eefb17a61c5587f56bb1750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-10 18:41:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e6be1fa361959aef92e3c82ccc507f64991d14f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:09:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77965084dafc4d30b557ba171c2d58db : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0a7c07863ee41d99e804e46b2c17bcb : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d77867a282c8baf324cb9ec5735fec2820fe313e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-10 19:52:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/6/iscsi/ : SUCCESS in 57m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/6/smb/ : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":6},{"id":"d6fca2de4318ec9757ceecd089f7eb70a43bed1c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-10 20:07:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/53/784453/6/check/dsvm-tempest-zadara-driver/dd9371b : SUCCESS in 1h 09m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0cf42691c1717912b1555b01a4b25cf1ff9cb7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 20:15:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fda6af4b4416410a9b263da20de262db : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c86b36e92eb4fbba9d740d8473b190e : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/963161e9edd54f4187324afcf872d037 : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca1269af768a4357902d31e26b13d86a : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8d96d5929a4f74ba21c55126ccb07b : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/811ddf414b614261968ef701e9c957d1 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/7110bbeffd2d465181881b16217dd5d8 : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fce44cf8b33746aeae1e37182feca8e6 : SUCCESS in 1h 03m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4fc71911d6524cd4b52668b4a376211d : SUCCESS in 11m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f775c9b06ef8412688358148d82a372c : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ad2f7a1d87c4be2acbf134836658547 : SUCCESS in 7m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cea3b21320dc4c66a1d3c702b3caa0bd : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6836713b0dfb4bf6847e57bcffb86549 : SUCCESS in 41m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/591b48e5e242442a9b7cdf9a720d3ebd : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c770154748d4c1ebbdd3739f00ef5a9 : SUCCESS in 1h 15m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f0716c85bf4643b98cab9cae6f0c044a : SUCCESS in 1h 26m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b381f89fc89840dcbfa26495df68d403 : SUCCESS in 1h 18m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ced7f075b0b454d810d44939352abec : SUCCESS in 1h 24m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/10970bf1e81d429d8845a4a3b794c049 : SUCCESS in 1h 12m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/01c54dc34b2f42ad9c001b476435d06b : SUCCESS in 1h 11m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/280d2bc690d547499bae0e6a8020e806 : SUCCESS in 1h 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95e6ba239d7b4fbb9c3e80bb866a59fc : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":6},{"id":"ad6d6b61bc894d197f7eab78cf1ae114bca494ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-10 20:40:32.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/6/check/tempest-dsvm-full-focal-py3/7c6f123/ : SUCCESS in 1h 43m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/6/check/devstack-plugin-ceph-dsvm-focal-py3/b743e36/ : FAILURE in 1h 02m 55s","accounts_in_message":[],"_revision_number":6},{"id":"a736e26ef8dcd94ae60c7cfe8ac90b735386bca8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-10 20:59:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/53/784453/6/check/DellEMC_PowerFlex/ec0b378/DellEMC_PowerFlex/75 : FAILURE in 43m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"11cc0b6343e6fb2acb05e09fb5a9c060c2fefc6d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-10 21:15:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/6/check/cinder-storpool-tempest/426257e/ : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":6},{"id":"dd26e8ea91cb3d0ad82b1f2dd6d02f61f4c0a903","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-10 21:20:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/6/upstream-solidfire-check/cinder-solidfire-iSCSI/489cbf5/ : FAILURE in 1h 02m 07s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"1e45e231f0114b9cdb443cf66a0a13e1d0d74f36","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-10 21:44:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/603465e/ : POST_FAILURE in 15m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/35713fc/ : FAILURE in 33m 31s","accounts_in_message":[],"_revision_number":6},{"id":"b51e061145620ce3fa36ac183e1d37e57084ebfe","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-10 21:57:56.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/53/784453/6/check/DellEMC_PowerStore_iSCSI/1adb651/DellEMC_PowerStore_iSCSI/67 : SUCCESS in 46m 33s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/53/784453/6/check/DellEMC_PowerStore_FC/12dc8f6/DellEMC_PowerStore_FC/68 : SUCCESS in 47m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"5540bd3d9cb1618f951ad0bf4b66efa315baf56c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-10 22:08:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/6/upstream-check/cinder-cDOT-iSCSI/1de0df6/ : FAILURE in 1h 19m 35s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/6/upstream-check/cinder-cDOT-NFS/f0366ba/ : SUCCESS in 35m 55s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/6/upstream-check/cinder-cDOT-FCP/7c9d7c2/ : FAILURE in 1h 50m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"4f336fed206181938a94194cec290829e790a27e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-10 22:18:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/6/check/iscsi-driver-primera/c839ad2/c839ad2.html : FAILURE in 58m 55s","accounts_in_message":[],"_revision_number":6},{"id":"29cd38c7514d437386679bd57234cc45643032c2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-10 22:42:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/6/check/dsvm-tempest-cisco-zonemanager-job/7f103b9/ : SUCCESS in 1h 12m 21s","accounts_in_message":[],"_revision_number":6},{"id":"be94063bd3d6a155b545d6cdd23b92dee661156f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 03:56:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d99249e/d99249e.html : FAILURE in 53m 44s","accounts_in_message":[],"_revision_number":6},{"id":"e7641ad2ff2083c5eb3263251a261e3b37eb14be","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-11 04:45:12.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-6/ns5_nfs : SUCCESS in 5h 45m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-6/ns5_iscsi : SUCCESS in 5h 32m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"59ffccc3f2c4497286b1580c8e83ba9bb3bc6552","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-11 05:25:46.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/53/784453/6/check/DellEMC_SC_FC/1c01f23/DellEMC_SC_FC/87 : FAILURE in 2h 03m 01s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/53/784453/6/check/DellEMC_SC_iSCSI/0dbd9e3/DellEMC_SC_iSCSI/84 : FAILURE in 2h 10m 31s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"e67909469911334037620925023cd70db33b7035","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 06:55:26.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/6/check/iscsi-driver-nimble/b213edd/b213edd.html : FAILURE in 25m 40s","accounts_in_message":[],"_revision_number":6},{"id":"2e05f13e9f356daa2cc3765d24690abe85c09f0a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-11 07:32:47.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-6/Seagate-iSCSI : SUCCESS in 45:35.05\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-6/Seagate-FC : SUCCESS in 38:26.10","accounts_in_message":[],"_revision_number":6},{"id":"3fe28edae05a62020b9e9db69fc36640ad8114c2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-11 08:59:39.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-6/Lenovo-iSCSI : SUCCESS in 48:17.35\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-6/Lenovo-FC : SUCCESS in 43:49.15","accounts_in_message":[],"_revision_number":6},{"id":"09d321db0245a9a075921f86a56c5dee537335e7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-11 09:41:08.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/6/check/DellEMC_Unity_FC/432ac5e/DellEMC_Unity_FC/150 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/6/check/DellEMC_Unity_iSCSI/e8f03c7/DellEMC_Unity_iSCSI/138 : SUCCESS in 1h 33m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ec551234c1dcf940fc4766907a234bf7c70dedcb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-11 13:08:31.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/53/784453/6/check/DellEMC_VNX_iSCSI/1e0ac6a/DellEMC_VNX_iSCSI/91 : SUCCESS in 3h 09m 37s\n- DellEMC_VNX_FC http://publiclogs.emc.com/53/784453/6/check/DellEMC_VNX_FC/6ead570/DellEMC_VNX_FC/95 : SUCCESS in 1h 23m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"8cd80e4fa23c626182f3c2a9881afb76f3fdde43","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-11 14:01:36.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-6/HPMSA-iSCSI : FAILURE in 1:26:43\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-6/HPMSA-FC : SUCCESS in 34:21.74","accounts_in_message":[],"_revision_number":6},{"id":"a698d0cdc8b643bc374027e49f7776f19265a9a8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-12 00:06:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/6/20210512062538/NEC-ISCSI : SUCCESS in 2h 33m 36s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/6/20210512062538/NEC-FC : SUCCESS in 2h 40m 47s","accounts_in_message":[],"_revision_number":6},{"id":"2a589262799fb5d0286558c4bfc6c2a97a334a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-30 20:59:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ef1ec3f4f3b14069487f3d9ed3d822043225e835","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:22:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d84f2d3a3a24a04af046fc893a94693 : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a001447c2db4e49af1a7998ea054745 : SUCCESS in 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6a0d4670af11238ed5a7d11e4a21f67a2fa49c63","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-30 21:51:00.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"ddd600eac8c7e0511b8d8142e8edc8108da62bc7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-30 22:06:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/7/iscsi/ : SUCCESS in 1h 05m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/7/smb/ : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c3f21c0986ed297a90665f3e844221a33cfd207e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-30 23:03:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/7/check/iscsi-driver-primera/40095ee/40095ee.html : FAILURE in 52m 54s","accounts_in_message":[],"_revision_number":7},{"id":"e95a64c1ca79282a8a9534d3ff0fe6792869838d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 23:29:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a430ed700ab747e395777b236e293d79 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff9c4cdb82814e7295985e1f7b09a695 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef867b569053406585b665f73305c2da : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bde0e81dd7c440a8b5a137ec3cbebdc : SUCCESS in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52ba447749f940c1aa6333c0487440f8 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2e13c6a2db47df8c3d7f709faeca56 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/7ba53d379e9744729af70ae5ab239b69 : SUCCESS in 1h 05m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e73978a6db754b649c4c0a0de5d0da28 : SUCCESS in 1h 08m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5c6ee7b64794fd68cefeff06ac0ee61 : SUCCESS in 13m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0572c58ee9dc41d197c5534c5cc0d564 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f5685b29cbf4ce89b7962b76451034d : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e9a5953d2ca4951b9115eb9ff4fca97 : SUCCESS in 9m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60a1f958b22d41ba85154531ec330e81 : FAILURE in 31m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4d2478d12f0e40c4b7c9937f1044a778 : FAILURE in 7m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b1e3e9878ad4459a7032164060164e6 : SUCCESS in 1h 19m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18ad9c58cf3b4a65a1ba5bfd013d9c78 : SUCCESS in 1h 35m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0165ba466130415c8290ff41609fd0b9 : SUCCESS in 1h 42m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/58475e581d2c4e5487e33ecefef8cbb4 : SUCCESS in 1h 14m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a87535bf88504f699aa020944f7bac13 : FAILURE in 1h 22m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b2448b8ea41493090474c69e9e3964e : SUCCESS in 1h 16m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b07c4dc9aa04f3799d0a1e2a3c46cc4 : SUCCESS in 2h 23m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e25e1ad426e47319ddea9167dc5c914 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":7},{"id":"55f7fbaa39e791efe3247e1cba0b6f35f9809bc8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-30 23:30:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fc9da3c/ : SUCCESS in 1h 19m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e714beb/ : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":7},{"id":"74121d0c3ac0f72c1d04a34a9a08cbcfb9261db5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-30 23:36:54.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a7a113eee18e4f9b8360e3e1d6490d1b : DISK_FULL in 1h 49m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9b6682804c4b4334a12a21f745b2ff77 : POST_FAILURE in 56m 53s","accounts_in_message":[],"_revision_number":7},{"id":"2a4809ddcfe9fb323d5306e8a44b650be0160c77","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-01 00:12:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/7/check/dsvm-tempest-cisco-zonemanager-job/3af6ddf/ : FAILURE in 1h 04m 07s","accounts_in_message":[],"_revision_number":7},{"id":"ded14900fc8c2cd34bb2d8c8578f72af409bfd99","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-01 00:26:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8ea0ecc/8ea0ecc.html : FAILURE in 51m 31s","accounts_in_message":[],"_revision_number":7},{"id":"e188d65fd4b462590e4676a2a0a7f92ed80b48ed","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-01 00:38:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/7/upstream-check/cinder-cDOT-iSCSI/7fbe5b9/ : SUCCESS in 34m 48s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/7/upstream-check/cinder-cDOT-NFS/6fcfac6/ : SUCCESS in 37m 12s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/7/upstream-check/cinder-cDOT-FCP/6f98060/ : SUCCESS in 1h 06m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f0928df66b69fb781b7f520250faccfa4fadd9dd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-01 01:32:00.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/53/784453/7/check/DellEMC_PowerStore_iSCSI/fb013dc/DellEMC_PowerStore_iSCSI/197 : SUCCESS in 55m 39s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/53/784453/7/check/DellEMC_PowerStore_FC/8acc367/DellEMC_PowerStore_FC/198 : SUCCESS in 51m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"a3fcdcfc7fb4f451bec1ca38fac58824e785ee39","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-01 02:22:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/7/upstream-solidfire-check/cinder-solidfire-iSCSI/1c909d0/ : FAILURE in 2h 50m 04s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"fcd9077bbe2782cae94332f564c353af8dcb1f1f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-01 03:20:40.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/13886/ : ABORTED in 25s","accounts_in_message":[],"_revision_number":7},{"id":"7ba330e35ffc27fcac584f74fd3e96c1685b5d96","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-01 03:52:00.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/53/784453/7/check/DellEMC_PowerFlex/cee902e/DellEMC_PowerFlex/266 : FAILURE in 1h 43m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"c094eec621b7870e67a7e7ff504ecc39bc09cbd4","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-01 04:54:28.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-7/Seagate-iSCSI : SUCCESS in 41:39.07\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-7/Seagate-FC : SUCCESS in 36:04.12","accounts_in_message":[],"_revision_number":7},{"id":"f791a49edf6617deb65fd434699b3d32a3b30214","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-01 05:07:26.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/7/check/DellEMC_Unity_FC/a4ca768/DellEMC_Unity_FC/388 : FAILURE in 1h 51m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/7/check/DellEMC_Unity_iSCSI/dfd1e5a/DellEMC_Unity_iSCSI/364 : FAILURE in 1h 17m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4c336dc2ea68f0c8cebb4eb49ea85e8d0d3bb615","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 05:08:38.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"f434e99aa17b68dca9cbe4823c481dc562c49292","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 06:12:35.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/7/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"a389de020eb6e5579991b215396fb20e853676f9","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-01 09:52:53.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-7/HPMSA-iSCSI : FAILURE in 58:28.24\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-7/HPMSA-FC : SUCCESS in 31:44.08","accounts_in_message":[],"_revision_number":7},{"id":"661d2fad4d8659e5675467f14b20a18cf384bcd5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-01 12:30:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/7/check/DellEMC_XtremIO_FC/24a6379/DellEMC_XtremIO_FC/431 : SUCCESS in 1h 23m 27s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/7/check/DellEMC_XtremIO_iSCSI/8794182/DellEMC_XtremIO_iSCSI/424 : FAILURE in 2h 58m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"259efede4f1bed344ba22520719d8a4b9924aece","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 15:12:04.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"bbd25c65cf8fd6c632c4b40762d88304b8913040","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-01 16:49:33.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-7/PVME-iSCSI : SUCCESS in 58:19.83\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-7/PVME-FC : SUCCESS in 52:52.06","accounts_in_message":[],"_revision_number":7},{"id":"ac0940c19c79628b8ac629cbf49708acd9dc0cd8","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-02 03:01:40.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-7/Lenovo-iSCSI : FAILURE in 3:52:19\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-7/Lenovo-FC : SUCCESS in 43:21.90","accounts_in_message":[],"_revision_number":7},{"id":"b4c93cfd3751943c2a9fb1cb69c334b19e542fb3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-02 11:05:16.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-7/ns5_nfs : SUCCESS in 4h 51m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-7/ns5_iscsi : SUCCESS in 4h 55m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"bcc82d30ce4f5d4405dae3b3ed80a19d5d19197e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 12:27:07.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"4eb09510b079cc1e96d5d5bf7246dcf9cb515489","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-02 16:46:02.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/53/784453/7/check/DellEMC_VNX_iSCSI/6951a9e/DellEMC_VNX_iSCSI/227 : SUCCESS in 3h 16m 09s\n- DellEMC_VNX_FC http://publiclogs.emc.com/53/784453/7/check/DellEMC_VNX_FC/67de29b/DellEMC_VNX_FC/245 : FAILURE in 1h 55m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"c40b4717412f5b1f8b000577da2100633719a5ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 18:14:54.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"eb5fab96f6b5852d349aa1830f1d79a6616a8303","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-02 21:14:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/7/20210703012656/NEC-ISCSI : SUCCESS in 2h 44m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/7/20210703012656/NEC-FC : SUCCESS in 4h 47m 43s","accounts_in_message":[],"_revision_number":7},{"id":"ac5485cef2437f09940f23a2bf93746e4fb1cac2","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-05 19:49:35.000000000","message":"Patch Set 7: Code-Review+1\n\nLooks good to me. I apologize because you probably already said this but I couldn\u0027t find any reference of the mypy-files.txt ,are we using it to track the mypy work of it\u0027s a config file?","accounts_in_message":[],"_revision_number":7},{"id":"48eab69ff89edd2ec58582100cad6bb458cc9f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-13 13:49:38.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"d1b7d4e24a53659762a92af7f209914daa5806e2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 13:55:03.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"0d47eb8033157ead23081d8a2ffe9a946064a1bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 14:13:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b19b279d9dc54f87b45607c6a34008f2 : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8a125d5c04045049cf762b46328b37b : FAILURE in 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"faa3550f7057d1c5c6cbf11527876c36e4e3c55c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-13 14:16:00.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"213ce299ff3b7fdd3adac6034abf67384f6ab404","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 14:27:21.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"6a1981c1d97fa9e73865d4a81d282a0f94a8972e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-13 14:45:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/8/check/iscsi-driver-primera/bdfb05b/bdfb05b.html : FAILURE in 53m 31s","accounts_in_message":[],"_revision_number":8},{"id":"ca8972d11c62fb244117c656e61e8a21e789410e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-13 14:45:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/26ae333/26ae333.html : FAILURE in 53m 56s","accounts_in_message":[],"_revision_number":8},{"id":"479bd3ccaa48bae1d5fef017af632d74ae87d4b2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-13 14:49:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/8/iscsi/ : SUCCESS in 58m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/8/smb/ : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":8},{"id":"310480eb24b70096bff5fea6bd763662068e51bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-13 15:26:10.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/8/check/DellEMC_Unity_FC/6e86650/DellEMC_Unity_FC/425 : SUCCESS in 1h 06m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/8/check/DellEMC_Unity_iSCSI/e4a12fa/DellEMC_Unity_iSCSI/400 : FAILURE in 1h 35m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"2ee0e4c7ed86bc93a35f393462264a8ee415a6c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-13 15:26:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4d92336/ : SUCCESS in 1h 35m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/516169f/ : SUCCESS in 1h 32m 37s","accounts_in_message":[],"_revision_number":8},{"id":"50881e9e5c2a032464a0debec732d9f49f8c8335","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-13 15:44:27.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/8/check/DellEMC_XtremIO_FC/352079d/DellEMC_XtremIO_FC/492 : SUCCESS in 1h 24m 39s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/8/check/DellEMC_XtremIO_iSCSI/50e68c8/DellEMC_XtremIO_iSCSI/482 : SUCCESS in 1h 54m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"fb63ee6497a61762056b77cb63407ae9fdd95185","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 15:45:03.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"862fbcf180b1840d161a74456ef95e4cfd71b842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 15:53:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9313b711486d4930a6c90505e90da871 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ce72094f5754584a20d16c13acb2d29 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4c353d43f164e6fa88052e4ac4ea4c8 : SUCCESS in 14m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/776169d826754e0dbf7bdace80a0ac5e : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5be747961b4089a5fb11db44b9871e : SUCCESS in 11m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/b82cf4e4f9f84691981377af72d5d620 : SUCCESS in 57m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fa9f46a2c8b4d789ab7075dd8fcea38 : SUCCESS in 1h 02m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84dd7e53d7aa4de4aabe505bccb03527 : SUCCESS in 10m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d941f4a031214ad5950ad7a80d44184f : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f28e23dc10154a09aa11c89fb4fc7bc2 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e34300b7d75f440fb3b408c938459f12 : SUCCESS in 7m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fa809affc6514d4fbadbe1ad8079e26f : FAILURE in 26m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a61a5bc591346348b391d11df0d4851 : FAILURE in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc2e3c6df42f48e8919a1a226f18ba5e : SUCCESS in 1h 09m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a079981d9ebd4e0b98bdeea3323a609a : SUCCESS in 1h 40m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af6320d165eb4e709506b5daf2e6ea40 : SUCCESS in 1h 18m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d68bcfb1dcc46099f710366e0d1fdbf : SUCCESS in 1h 12m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8674d280c40b4f29b2b94856bd205171 : SUCCESS in 1h 07m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/649e8cbe6b954c64adffe6fe1b607646 : SUCCESS in 1h 15m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1756d27db8d54e2c8dd0ae557eefe7d1 : SUCCESS in 2h 01m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a848bdf9e904a70b1c44f59a79d1742 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":8},{"id":"e2f28ec300e42207d5d4e0c86531912ef7aee306","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 16:13:51.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"18a069d329a443ea193a202875b94aae65931aab","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-13 17:20:31.000000000","message":"Patch Set 8:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-8/PVME-iSCSI : FAILURE in 9:26.38\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-8/PVME-FC : SUCCESS in 47:06.35","accounts_in_message":[],"_revision_number":8},{"id":"41e7a198cc3d69b71c330c1b055093a7515e9d15","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-13 17:22:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/8/upstream-check/cinder-cDOT-iSCSI/ddf7490/ : SUCCESS in 44m 19s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/8/upstream-check/cinder-cDOT-NFS/6651b65/ : SUCCESS in 43m 31s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/8/upstream-check/cinder-cDOT-FCP/d5b8b31/ : SUCCESS in 1h 27m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"e80ac5de26e2189b263385feba3f5e3d0ffdb134","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-13 17:46:04.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-8/Seagate-iSCSI : SUCCESS in 43:17.40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-8/Seagate-FC : SUCCESS in 39:00.46","accounts_in_message":[],"_revision_number":8},{"id":"2f5a5b16ea93fad85881b4b66ce669287a25501b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-13 17:46:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/8/upstream-solidfire-check/cinder-solidfire-iSCSI/ed526c2/ : FAILURE in 1h 51m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"3eaf77230506e6685274f30fc5773e3f99109377","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-13 17:58:14.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-8/HPMSA-iSCSI : SUCCESS in 1:00:04\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-8/HPMSA-FC : SUCCESS in 33:35.82","accounts_in_message":[],"_revision_number":8},{"id":"576519ebf4fa59428e70dbd098d29ec065499d40","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-13 18:21:00.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-8/ns5_nfs : SUCCESS in 4h 27m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-8/ns5_iscsi : SUCCESS in 4h 30m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"811d46d4a7ed56bffa669a0598563ece011e2e7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 18:51:45.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"e2d57250a177f100b35b5986a687cf6ed49d18eb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-13 18:51:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/53/784453/8/ : SUCCESS in 2h 26m 25s","accounts_in_message":[],"_revision_number":8},{"id":"0870fb45b770fb79a33f0805811a54776bc16047","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-13 20:17:52.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-8/Lenovo-iSCSI : SUCCESS in 1:41:24\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-8/Lenovo-FC : SUCCESS in 2:13:21","accounts_in_message":[],"_revision_number":8},{"id":"0a5ef0bafc0037ef917850668f7df58cabda0306","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-14 03:56:49.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/8/check/DellEMC_Unity_FC/b1c1eae/DellEMC_Unity_FC/432 : SUCCESS in 1h 09m 40s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/8/check/DellEMC_Unity_iSCSI/cbe0780/DellEMC_Unity_iSCSI/407 : SUCCESS in 1h 47m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"5a9d1c5895fff1c76d530ecf0ef539fe324d0be3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-14 05:47:02.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/8/check/DellEMC_XtremIO_FC/3e5035b/DellEMC_XtremIO_FC/499 : FAILURE in 1h 16m 31s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/8/check/DellEMC_XtremIO_iSCSI/5c18799/DellEMC_XtremIO_iSCSI/489 : FAILURE in 2h 24m 52s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"f853d4f7d06961ef929515b127cee02215263149","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 05:47:42.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"76890485d605a1a24dd7785309289b172824b2bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 07:15:29.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"97948200ddccecc03b1724c01e929adc4c8b64e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-15 17:36:36.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"55dd53a6a18ba775a0cf9d0ffddf47775109e8e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:56:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48b4d3f32ec1407aa73d87d04d340740 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75483d01a9a74facaf1520b24de7d2a7 : SUCCESS in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8b5be397dcc25cc9fa30950ca131c3b0cdee6b40","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 18:01:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/8/20210716002149/NEC-FC : SUCCESS in 2h 32m 43s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/8/20210716002149/NEC-ISCSI : SUCCESS in 2h 39m 12s","accounts_in_message":[],"_revision_number":8},{"id":"5c4d6c5da1f261eb315415d4bc3037e24b1e46cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 18:02:47.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/9/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"74ff60e28a3d4b871cb93bc6508c775db19a3c30","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-15 18:27:53.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"2b5b8bbd042cb32e11a8a713443e0174808ed8b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 18:29:06.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/9/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"3de521c46aba0180f8b5ab2fe9fa00799699e268","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-15 19:25:33.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/9/iscsi/ : SUCCESS in 1h 14m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/9/smb/ : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":9},{"id":"8933c1e7f8997b072092db396b1aa07aa8177eda","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 19:39:36.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/9/check/iscsi-driver-primera/28de8cc/28de8cc.html : FAILURE in 55m 38s","accounts_in_message":[],"_revision_number":9},{"id":"5143a358d1d46f1b316725f132993d595e9508fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:43:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0e50e057c9440b8ddddf40271304ce : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16ca0312512945c7aa9f45c13f85c4da : SUCCESS in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae6fb88d32ab4bd8a77b2f36a5f084b8 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8979c3b8c9545059f3039594e4880f8 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85a259441a140c18217587793d9d213 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/8e3e50b277284217ad289d0a6f10d5f2 : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffd13df73f3f4422aa901e2199fd356e : SUCCESS in 1h 02m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba08811f2e3c4d629d478894cc97a1fc : SUCCESS in 12m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e14eb0bb5db348609879cac2a9c2b569 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ca8d9dfb460407582b72e5b9aea7a51 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/208149e18d1f4f46874f5ef5610e054d : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2a2f56a1f9d495aa7a1e0ad119d1787 : FAILURE in 31m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9b0e8065bcb45798f75addd9d855aed : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f8387fcf71c54e6d9b6e5390d4ef975e : SUCCESS in 1h 15m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08c415b7c6504ff2831313b57888f387 : SUCCESS in 1h 21m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7102dc268b1b44dea8fc1bf115d04f6f : SUCCESS in 1h 21m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9135d09ba59f49d181b1e4be4bcca69a : SUCCESS in 1h 26m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd8fed82d9fa449fb9272d66254c61ae : SUCCESS in 1h 09m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3d2270d890e400db779bd2dfcbb362c : SUCCESS in 1h 15m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7747ff2c6a4a42d58c47df17eb8d7924 : SUCCESS in 2h 04m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35e0f476502e41d98562c65a680368c0 : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":9},{"id":"b03608f5f52f6a41f6051c31156e368359baf031","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-15 20:13:35.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/784453/784453/9/check-not-vote/ext-cinder-zuul/9176380 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":9},{"id":"b38abe9aa225504a5cf17d6915360fb066185fe0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-15 20:30:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/9/upstream-check/cinder-cDOT-iSCSI/ad154d4/ : SUCCESS in 44m 37s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/9/upstream-check/cinder-cDOT-NFS/1af6004/ : SUCCESS in 45m 14s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/9/upstream-check/cinder-cDOT-FCP/1dccf29/ : FAILURE in 8m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"ed842c92703e09861a3b83862ce77c707306f2df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 20:46:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c7fd4cb/c7fd4cb.html : FAILURE in 52m 10s","accounts_in_message":[],"_revision_number":9},{"id":"6c805a56b527e726161c3c95bbc411caf1b9d7b9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-15 21:12:02.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b362e8b/ : SUCCESS in 1h 25m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/137b291/ : TIMED_OUT in 3h 05m 02s","accounts_in_message":[],"_revision_number":9},{"id":"905c7acfbed01afdd9ecf5694774138b7053d972","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-15 21:12:05.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/53/784453/9/check/DellEMC_PowerStore_iSCSI/8006511/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/53/784453/9/check/DellEMC_PowerStore_FC/31c0cc5/DellEMC_PowerStore_FC/246 : SUCCESS in 48m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"952559aee6be28860ee8a734204b8e2df796447e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-15 21:35:20.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/9/upstream-solidfire-check/cinder-solidfire-iSCSI/8b68920/ : FAILURE in 1h 49m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"03a59fd24d088ae31299f20f05827cae177fceb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 22:15:55.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/9/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"d640496ea29aaa70b3472a5c0cb93dceb9ec4aee","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-15 22:18:37.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/9/check/DellEMC_Unity_FC/e86e4ca/DellEMC_Unity_FC/471 : FAILURE in 1h 16m 37s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/9/check/DellEMC_Unity_iSCSI/25dc1a1/DellEMC_Unity_iSCSI/442 : FAILURE in 1h 13m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"85ab9fe629c35d6ede2aeb1096fbf895fcc7785a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-16 00:08:42.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-9/ns5_nfs : SUCCESS in 4h 50m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-9/ns5_iscsi : SUCCESS in 4h 56m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"ff95230519c13ec55bd47489e388559c39a85a51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 00:11:18.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/9/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"3cd59f59944fb6808ee88db46e397baf583e0c5b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-16 00:21:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/53/784453/9/ : SUCCESS in 2h 08m 17s","accounts_in_message":[],"_revision_number":9},{"id":"9cf750faa7008dc31a08bc13b93e3a740f43bd6f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 00:33:22.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/53/784453/9/check/DellEMC_XtremIO_FC/9514ea6/DellEMC_XtremIO_FC/535 : SUCCESS in 1h 15m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/53/784453/9/check/DellEMC_XtremIO_iSCSI/9196ec3/DellEMC_XtremIO_iSCSI/521 : FAILURE in 1h 46m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"2a825dfd89fe7e0f6b4391add5df8e93abf37fa2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 01:41:04.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/9/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"2591e1d9f83db7a274fb78b01264d9d725ee1492","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-16 02:49:35.000000000","message":"Patch Set 9:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-9/Seagate-iSCSI : SUCCESS in 44:57.87\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-9/Seagate-FC : SUCCESS in 38:53.42","accounts_in_message":[],"_revision_number":9},{"id":"bfee040da219974c11b63ed32586e0607e9d3eb0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-16 03:25:10.000000000","message":"Patch Set 9:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-9/HPMSA-iSCSI : FAILURE in 59:47.49\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-9/HPMSA-FC : SUCCESS in 33:31.21","accounts_in_message":[],"_revision_number":9},{"id":"ef264efe0690a725df1af8acd07a92d248acf78e","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-16 10:19:31.000000000","message":"Patch Set 9:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-9/PVME-iSCSI : FAILURE in 44:05.77\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-9/PVME-FC : SUCCESS in 47:02.01","accounts_in_message":[],"_revision_number":9},{"id":"f74429e7a317e57295eda639c1cfeba0e34cf5cf","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-16 19:52:37.000000000","message":"Patch Set 9:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-9/Lenovo-iSCSI : SUCCESS in 51:05.44\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-9/Lenovo-FC : SUCCESS in 48:14.22","accounts_in_message":[],"_revision_number":9},{"id":"fba2ccbd32097e5ed0a38e3efe110ea6f378a536","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-17 12:13:00.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/53/784453/9/check/DellEMC_PowerMAX_FC/09d69bf/DellEMC_PowerMAX_FC/265 : SUCCESS in 5h 42m 58s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/53/784453/9/check/DellEMC_PowerMAX_iSCSI/e88f491/DellEMC_PowerMAX_iSCSI/251 : FAILURE in 5h 42m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":9},{"id":"9e79a0651b29ea012643caae4fef6a53082cb8b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-17 13:28:18.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/9/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"31e03b9b570b2dc10e5b0f03601c5b9a91b8a11a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-18 14:54:31.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/9/20210718211736/NEC-ISCSI : SUCCESS in 2h 30m 39s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/9/20210718211736/NEC-FC : SUCCESS in 2h 36m 50s","accounts_in_message":[],"_revision_number":9},{"id":"c0a6625fdc2abdc4398e121b9af3e9c26b02a632","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-28 16:00:42.000000000","message":"Patch Set 8:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/53/784453/8/fc : SUCCESS in 1h 05m 43s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/53/784453/8/iscsi : SUCCESS in 0h 53m 19s","accounts_in_message":[],"_revision_number":8},{"id":"8b618ecab637bc216b0a5a795fe321b6fe019bc7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-28 16:00:45.000000000","message":"Patch Set 9:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/53/784453/9/fc : SUCCESS in 1h 05m 48s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/53/784453/9/iscsi : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":9},{"id":"39cc0d555fcd76276d165fc7038d546fc964cb8e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-04 16:06:46.000000000","message":"Patch Set 9: Code-Review+2\n\n+2 I\u0027ve already reviewed this and looks good to me","accounts_in_message":[],"_revision_number":9},{"id":"02225715d2fbdab940f76aba30d03c4e592cf8cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-09 13:09:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b7be9ee18870242b5c57d7e3239e883adf1f5c7a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-09 13:16:04.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/10/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":10},{"id":"53877bc061b286eb15baf0241e7518c0dc2c0e05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 13:36:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e762a7425594dc3ae56a8b464732547 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ec47cf32f134cc1a9432c1f688296f1 : SUCCESS in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f0d260f1222f6edd5f45a89a063dd063525a2e18","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-09 15:06:50.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/10/iscsi/ : SUCCESS in 54m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/10/smb/ : FAILURE in 12m 59s","accounts_in_message":[],"_revision_number":10},{"id":"652dafd4e136220b9ee763e062b4e8f9dd106a48","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-09 15:07:39.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/10/check/devstack-plugin-ceph-dsvm-focal-py3/758c15a/ : FAILURE in 59m 18s","accounts_in_message":[],"_revision_number":10},{"id":"e6c4435b0763f872d986db1d8608f0f16698ca8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:13:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22ff78111f446ec9443a31e6b22182f : SUCCESS in 10m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ab2557f2cc499e85ef26f33483839b : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15e704a7d37e45eeb68a93139ddf3ada : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/795fefef6e0348e2bd90dab122c2c54e : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b815f854bc8c46d887ea79016f14674f : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/feabfcbb6fab45ddbb9b80bbbab1ed6c : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85c83976265f44ce898c39a50a7d7b49 : SUCCESS in 1h 07m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4519dbc6cffe486183a411fb2bd89ea2 : SUCCESS in 13m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2f46a6b5aa834acab2f50c451197cf8b : SUCCESS in 6m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc6d1c63bbbf458a83db4e2599d52157 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b7a6285a2384f04be60e262147e32c2 : SUCCESS in 7m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/80a65598b4c243c697e3e386bdd57ba5 : SUCCESS in 9m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea53aefd974b4e8d860106ae96141d54 : SUCCESS in 37m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d1f7e883f8e45d08ec51b1cbf82419d : FAILURE in 7m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1fea7852917a47c0800d15f98047cf8a : SUCCESS in 1h 20m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c019bcf4e3a34665b411ab4ce4c0f201 : SUCCESS in 1h 23m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9ffb627022640678918705602b77c6d : SUCCESS in 1h 31m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/54de8138c7df4b0c9a43f7a58a5fa861 : SUCCESS in 1h 14m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bdbdc58a27274ca3a6a965f45739d0bb : SUCCESS in 1h 21m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/65f77f20bb764f6988d7a79ea35c3973 : SUCCESS in 1h 15m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/074579d60e634cfdaf898bc2c4df5179 : SUCCESS in 2h 02m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f85ebacb95e24949897aeaf5d4084985 : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":10},{"id":"943ab3934d41d20a553c9a97f617a7ed2c6e481b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-09 15:39:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"165b994d7cd6e6828adc95e123531edf4c01529d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:03:19.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b91c0c931db8430f97d44505b0052c1f : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f842e232d32a47a3b2a8ee5dd39860f6 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1c37daeba6ce6eda8561f7c3da63a756dcaef3ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:12.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/10/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"c4f4257eb66bab792eab4ed5d30de0ea92596c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 17:39:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eafea377afd84b8cb26ee4a7d4ef9284 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d49575957d4b66b959fd8888451c85 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfc400d77a5f4cb39ae23345c302cf9d : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5fb7e57424145edb7ed769284d608f3 : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f8239d510c4982b4a2882a92d47d0b : SUCCESS in 11m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/7d49bc44c6c04265b4a62586b3f70a51 : SUCCESS in 1h 03m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a0473dc82994d69b3e68e2afec0f285 : SUCCESS in 1h 04m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f64337462be0471abd8c1bc83df3e763 : SUCCESS in 13m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3c21f3b89566498c8c386b201666cb4f : SUCCESS in 5m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9401696d29fe43ebafa1cea037d63645 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a945a72e75124a5eb3144c0f653d6173 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/02201549cb6a48969ed863a8637a52e2 : SUCCESS in 7m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/17dda698584845729ace9b9899fdd705 : SUCCESS in 37m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/140553edf80a4f6bb62f602baddf8def : FAILURE in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f4fca01350a45ca92e1b0313dbaade1 : SUCCESS in 1h 17m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2795739a55b84db18f688f898638a8de : SUCCESS in 1h 30m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28690eab526c433494bbd470d719dd76 : SUCCESS in 1h 31m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f65a7966d110400d9f4946f832ecf374 : SUCCESS in 1h 07m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aebae25d335341a6b60bb468eadf8613 : SUCCESS in 1h 11m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/481730c1b1ec4f72b2c3d17c6bdaffa4 : SUCCESS in 1h 12m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/560f117e30f14e578aa66d6f3b63d72b : SUCCESS in 1h 57m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6678e90c85fb4cbb8273db399c386839 : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":11},{"id":"8dcecc7c6b981b1af80de8a54ba4e976f347356a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-09 17:57:27.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/11/check/devstack-plugin-ceph-dsvm-focal-py3/2946894/ : FAILURE in 1h 11m 54s","accounts_in_message":[],"_revision_number":11},{"id":"98c98bf3f48e5450fc3c61ad5ebb754367d668c1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-09 18:57:28.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/11/iscsi/ : SUCCESS in 54m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/11/smb/ : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":11},{"id":"8776666d7fc3d22403c789155c6046312f90e154","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-09 19:08:44.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/11/check/iscsi-driver-primera/ebaf027/ebaf027.html : FAILURE in 54m 23s","accounts_in_message":[],"_revision_number":11},{"id":"30686ed19d19f042c6de1510da2552f6c2271b02","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-09 19:27:33.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/11/upstream-solidfire-check/cinder-solidfire-iSCSI/f2bd22e/ : FAILURE in 1h 46m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"3cb2e934a646eb634def96e3ea99226e486bf342","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-09 19:42:39.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5ecc5ed/ : SUCCESS in 1h 27m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/11/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5fb1eef/ : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":11},{"id":"80c3c13a77e25a1753f1bbd7d55f3a22cd3fb4cd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-09 20:39:23.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/11/check/cinder-storpool-tempest/5f08828/ : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":11},{"id":"7a8cd25a61f730d2a7a6eadc8e3d322715c45711","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-09 20:52:32.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/11/upstream-check/cinder-cDOT-iSCSI/551c0e9/ : SUCCESS in 39m 30s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/11/upstream-check/cinder-cDOT-NFS/8421f45/ : SUCCESS in 39m 19s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/11/upstream-check/cinder-cDOT-FCP/5598480/ : FAILURE in 3h 12m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"405e516a35eab15db6ddcc907696e09e1450dabc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-09 23:09:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/11/check/dsvm-tempest-cisco-zonemanager-job/cf6610b/ : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":11},{"id":"ad37d49d3fba98ceda703f089f2db8fcec9d6440","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-10 01:05:53.000000000","message":"Patch Set 10:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-10/PVME-iSCSI : FAILURE in 14:46.81\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-10/PVME-FC : SUCCESS in 47:18.93","accounts_in_message":[],"_revision_number":10},{"id":"51e6f438ec8163ea2257a4724fd4d836e46326d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-10 01:29:57.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2f45a27/2f45a27.html : FAILURE in 52m 47s","accounts_in_message":[],"_revision_number":11},{"id":"7025176b4437087e6d9c30b1d1d38ffb54d58e88","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 01:50:26.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-10/ns5_nfs : SUCCESS in 6h 33m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-10/ns5_iscsi : SUCCESS in 6h 7m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"065d96267fdf18201170802a5e6de702e15eb596","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 02:13:05.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-11/ns5_nfs : FAILURE in 1h 5m 30s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-11/ns5_iscsi : FAILURE in 1h 1m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"30a82d5788ce15b1a73ca8293a27c0485b211f87","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-10 04:19:09.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/53/784453/11/check/DellEMC_SC_FC/5f477ba/DellEMC_SC_FC/267 : SUCCESS in 51m 36s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/53/784453/11/check/DellEMC_SC_iSCSI/9664a32/DellEMC_SC_iSCSI/263 : SUCCESS in 1h 48m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"21e76bfe52a5fde60f2d878fbbe6ac883ae7c2c3","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-10 04:56:06.000000000","message":"Patch Set 10:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-10/Seagate-iSCSI : SUCCESS in 43:11.03\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-10/Seagate-FC : SUCCESS in 42:16.11","accounts_in_message":[],"_revision_number":10},{"id":"2a8a74d15ae9984f9d1149a7a978fdaa96027513","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-10 07:06:17.000000000","message":"Patch Set 10:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-10/HPMSA-iSCSI : FAILURE in 57:54.23\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-10/HPMSA-FC : SUCCESS in 38:07.42","accounts_in_message":[],"_revision_number":10},{"id":"df1dc231b5a215a564b42ef4429897d0e541e23b","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-10 09:06:55.000000000","message":"Patch Set 11:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-11/PVME-iSCSI : FAILURE in 14:50.74\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-11/PVME-FC : SUCCESS in 46:09.66","accounts_in_message":[],"_revision_number":11},{"id":"8429271faae4f42539a60795f13d4cda138cce38","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-10 12:54:41.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/53/784453/11/check/DellEMC_PowerFlex/44612f2/DellEMC_PowerFlex/368 : ABORTED in 3h 00m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"10bac1c8820a7f6f6e0d4c51de87cd26f6f79142","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-10 13:02:17.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-10/Lenovo-iSCSI : SUCCESS in 2:16:15\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-10/Lenovo-FC : SUCCESS in 3:14:05","accounts_in_message":[],"_revision_number":10},{"id":"7fbb42e791465f7ebeb60b0c370f30438c5f2b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 13:24:58.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"d71ea2424e6400224e3e4467c47067271968774a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-10 13:27:51.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nQuestion inline, but otherwise I think this looks good.","accounts_in_message":[],"_revision_number":12},{"id":"ec44cb032d9736eda3708438727d49ebc55bd1d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:54:33.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c93e0c3694a4d978fc26ec26247ee1a : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c233f8cea3dc455aa711d07e87f3403e : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9cf878ef608bc30f70ac9dea403fa1e368ee6d0a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 13:59:20.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bdb843e9b9b9752b5e4f0c365baa4c5359eabedf","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-10 14:05:00.000000000","message":"Patch Set 11:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-11/Seagate-iSCSI : SUCCESS in 41:22.73\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-11/Seagate-FC : SUCCESS in 39:38.08","accounts_in_message":[],"_revision_number":11},{"id":"4e8d405e5bc35bf47305e5d1e77ab6fc71574456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 14:27:34.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8a542ce591a4b2a5b201090466af59268cc5e0e1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 14:42:38.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"61d0c75971a0f6d8d452dce18c067f70b97ba56b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-10 14:43:09.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"2c6c930cc267a3bae02bfdc9f3a09fdf5602492a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 14:51:52.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ac8e5a509b2651d77fefc700ce35d355dcd7ead2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 14:52:05.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"8dcdc906238a982223a3f9e084683225e6e1a6ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:16:44.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d091bea0292d4c7daa4aec57398477d1 : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c79cfcf82f94f08b02dca73ea9320a7 : SUCCESS in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4d5c4d294a7e7b38adbf0c37c51914b55c5bd668","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-10 16:14:02.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/14/iscsi/ : FAILURE in 1h 20m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/14/smb/ : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":14},{"id":"43f757ed94a73e5b48de2e18f35784cca93e91e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 17:12:08.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5057280d599746ffa47e1434202b3988 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda36f7334254345b1e61c1861204f7d : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99726abb8dab4607931d884d6f1ea63d : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8c3f1bd310d4226814e4b8ee300f105 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb40e1a020c94080bf466bb2301ff8cd : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/14910503a49b41abbca2136d323478a2 : SUCCESS in 1h 24m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e8cf419f0d444578b4e11b0f38f205b : SUCCESS in 1h 06m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a919d1ce924245b7bbc6ccf337a13af6 : SUCCESS in 12m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2458131d79f4dc699c76c92bf41e9d6 : SUCCESS in 6m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e0a942e5e91f48d2a5df3354351ba42d : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6a4ae7895394760a8cd5c924bb67184 : SUCCESS in 8m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb3f244de3954897a26a9ba2345b6f81 : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99b854929d1546acb0643d2cfe8d7718 : SUCCESS in 39m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0bb4fd41e1f04fc0bd97745cac6c604a : FAILURE in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea71c007cb9b4f039887e90a25ae70c5 : SUCCESS in 1h 19m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/30927da6db404dec832eb2f61031f11a : SUCCESS in 1h 45m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f54334c10f647dc9077fd2f08dd1dc3 : SUCCESS in 1h 21m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e5a1626f3774526b6e903144879bc5d : SUCCESS in 1h 12m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8fcea23bd49457c8817fc0bcc1a3555 : SUCCESS in 1h 40m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fd78122bfc624548a758879a889a8fee : SUCCESS in 1h 16m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21256409dc92451da9b05450f508474c : SUCCESS in 2h 15m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99ad7b3a8da14f65a596377a1e2afa73 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":14},{"id":"b87a8231fffdfb057bd9330edc2ca63ab985499c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-10 17:21:26.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/14/upstream-solidfire-check/cinder-solidfire-iSCSI/9223996/ : FAILURE in 7m 38s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":14},{"id":"687a0b842abcfbff987e32bb5cdb59dccfa3c960","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-10 17:41:53.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/14/check/cinder-storpool-tempest/907c2cd/ : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":14},{"id":"27bda6dd93f8be9c0e23ad0649e509a112321b47","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-10 17:48:03.000000000","message":"Patch Set 11:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-11/HPMSA-iSCSI : FAILURE in 1:00:27\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-11/HPMSA-FC : SUCCESS in 33:35.47","accounts_in_message":[],"_revision_number":11},{"id":"f24d011cbb66b8c6bf312768274f470ade3d1430","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-10 18:20:33.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/14/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/996247e/ : SUCCESS in 1h 26m 06s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/14/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0d974c6/ : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":14},{"id":"4d240db02d0da12bf0270cab59bd8aa1cfa8a2bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-10 19:29:30.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/14/check/iscsi-driver-primera/50520ec/50520ec.html : FAILURE in 23m 32s","accounts_in_message":[],"_revision_number":14},{"id":"7dd0802c8fdc61ba72c27f7895814d4742f6e8b4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-10 20:37:39.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/14/upstream-check/cinder-cDOT-iSCSI/2016947/ : SUCCESS in 3h 24m 28s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/14/upstream-check/cinder-cDOT-NFS/939b914/ : FAILURE in 2h 47m 51s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/14/upstream-check/cinder-cDOT-FCP/392b3ce/ : FAILURE in 2h 52m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"a476750f3f43d946b781aea824dbf798330fd743","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-10 20:40:47.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/14/check/dsvm-tempest-cisco-zonemanager-job/c753739/ : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":14},{"id":"baadf8cb23454e798d9b30454729797b5a61dcbb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-10 21:14:50.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)\n\nChanges LGTM.  One question inline.","accounts_in_message":[],"_revision_number":14},{"id":"feb148e85080b0a63845f4dce6a4d7ec63473f85","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 22:40:11.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-12/ns5_nfs : SUCCESS in 5h 58m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-12/ns5_iscsi : SUCCESS in 5h 58m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"07a141f9074bc59bb6c20c6420e18e4e52915818","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-11 00:05:56.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-13/ns5_nfs : SUCCESS in 4h 8m 6s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-13/ns5_iscsi : SUCCESS in 4h 12m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"5f22df61cff1e48aed0791161e990e1013b27fa3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-11 02:26:43.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784453-11/Lenovo-iSCSI : SUCCESS in 54:31.05\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784453-11/Lenovo-FC : SUCCESS in 51:15.77","accounts_in_message":[],"_revision_number":11},{"id":"ed71d5cc60d927673f1166447971ab5cb120ae5f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-11 03:06:00.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-14/ns5_nfs : SUCCESS in 4h 20m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-14/ns5_iscsi : SUCCESS in 4h 25m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"6c702bc64f62a8a6487303550f85cc7b4c2059fb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 03:47:35.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/7160635/7160635.html : FAILURE in 55m 31s","accounts_in_message":[],"_revision_number":14},{"id":"ebf9e5a3de93edf9dda9adda8a6c1fcfe64d6f1e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-11 05:54:15.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/784453/784453/14/check-not-vote/ext-cinder-zuul/eaae0c5 : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":14},{"id":"71342d0b12daffc34cdcaa002f7f6d3102ec7ef6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-11 06:26:26.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/14/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"15b11ddb07fab330f58c0fb61f62922dcd5dc4d0","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-08-11 07:20:00.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"1387bf0ed4c7e9ff776cd975631a50f577dcee48","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 08:44:17.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/14/check/iscsi-driver-nimble/4e8d68a/4e8d68a.html : FAILURE in 1h 58m 26s","accounts_in_message":[],"_revision_number":14},{"id":"7d5b99c2139670522a10c6c33f60a9fde0758c07","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-11 12:03:16.000000000","message":"Patch Set 14:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-14/PVME-iSCSI : FAILURE in 14:45.83\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-14/PVME-FC : SUCCESS in 47:46.77","accounts_in_message":[],"_revision_number":14},{"id":"a6bde11ad21a4f806c4c4392056ca25245158746","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-11 12:40:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"bad0784d290fe38653cfd15dd22f26944b9be5f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-11 12:41:15.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"10d788a798c493a3618179f0ca1b26bace069fef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 13:02:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3055375932f743658575109335c4464b : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e2e0766bfc64705a18c6ad6657d875f : SUCCESS in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5aa1b60ecfd53ead09ee92a2717e73f534347793","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-11 13:17:52.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/15/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":15},{"id":"79d79286f4f4ffa7fff2312ffe7d7afc053d9c01","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 13:35:36.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/53/784453/15/check/iscsi-driver-primera/094a171/094a171.html : FAILURE in 52m 11s","accounts_in_message":[],"_revision_number":15},{"id":"49c40293feee49023f93a5819af4b37eea122716","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-11 13:44:50.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784453/15/iscsi/ : SUCCESS in 1h 02m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/784453/15/smb/ : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":15},{"id":"e9b4772ad3ba8f399f5cdb14771c8b8fa9a82505","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-11 14:03:18.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/53/784453/15/check/cinder-storpool-tempest/f43daf5/ : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":15},{"id":"d2d0179114862d0593254e5af459d9a4ee046f21","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 14:09:52.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/53/784453/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/eb96f45/eb96f45.html : FAILURE in 51m 29s","accounts_in_message":[],"_revision_number":15},{"id":"c36ea43b898c68f5aa99dde03b67f29517e67b7d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-11 14:31:23.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/53/784453/15/check/dsvm-tempest-cisco-zonemanager-job/247a1de/ : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":15},{"id":"8c5062b9bb5a8e5d9c7b7f02fd38283179b2fa74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 15:16:10.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f064ddf7ac4d74acb718399575e973 : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea8b8c2af6346c0b4925a76b14b2905 : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00ac110782a746d8a286922df833301f : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3959d9443a5342a5aab8a92363827d7e : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89375082e1934c1ea09e68f881302c70 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/af0fd288badb48f6a723b0252cb941e2 : SUCCESS in 1h 01m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2e0df0553e2455886eed8f0881b105b : SUCCESS in 1h 00m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fbdd0922ca7f4303978b32c1bb492ffc : SUCCESS in 12m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/310d29d1a362426e8a98ec67392c2083 : SUCCESS in 6m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e738001fc73a4cfd951a28278f630058 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae8663d3c9f64b38b5b283c3a4317b24 : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e79d4a063a743f88f81f893dd103d8d : SUCCESS in 7m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/641f101859a7489cb12b9938f0851ace : SUCCESS in 35m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19e9a43534064b1995718d2bf5540def : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec82393788a84503b83115d02cee2ce4 : SUCCESS in 1h 17m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/13bc64626e3647d89ec745b9ea9a3f08 : SUCCESS in 1h 30m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90e6a469621e49afaef70b2ac2e610e1 : SUCCESS in 1h 24m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f46fc60bc99e49a9b24c1be556953cf1 : SUCCESS in 1h 09m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eda00983329b47b2ab7f14494b90edbb : SUCCESS in 1h 12m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab30f7dffc91454eb06fa738a1f8c2b3 : SUCCESS in 1h 26m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46fcac5007e949bab172b0b7adbce207 : SUCCESS in 2h 28m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d65f8217b4ea4ec3a31adbca788f9704 : SUCCESS in 45m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a52f804045c54c0ba5e5d789c355ad17 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":15},{"id":"98d081321bf931752fba5b21d2b3584a82178568","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-11 15:45:28.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/784453/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d7fc836/ : TIMED_OUT in 3h 03m 02s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/784453/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/be46ce4/ : TIMED_OUT in 3h 03m 03s","accounts_in_message":[],"_revision_number":15},{"id":"34553a2c7d16d37a41c6261621ce3dc6ccbb0800","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-11 16:59:43.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-15/ns5_nfs : SUCCESS in 4h 13m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-784453-15/ns5_iscsi : SUCCESS in 4h 17m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"38ad66b325ee0cc23f43ac75305cf20024db17a0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-11 17:00:19.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/53/784453/15/upstream-solidfire-check/cinder-solidfire-iSCSI/f4cf12f/ : FAILURE in 1h 42m 35s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"29253ab571517b7acc8c98475d17a566dcaba87c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-11 17:38:11.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/53/784453/15/upstream-check/cinder-cDOT-iSCSI/66463ff/ : SUCCESS in 35m 43s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/53/784453/15/upstream-check/cinder-cDOT-NFS/c4dc85e/ : SUCCESS in 37m 51s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/53/784453/15/upstream-check/cinder-cDOT-FCP/0bb6eef/ : FAILURE in 2h 20m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"f44e527340a45d296cf059bb855e338c21043f3f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-11 18:24:06.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/53/784453/15/check/tempest-dsvm-full-focal-py3/36e00e4/ : FAILURE in 2h 19m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"6945d3837d466a46e8ebe68d37f61b36c5af963d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-11 19:57:37.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/53/784453/15/check/DellEMC_Unity_FC/56f3f9d/DellEMC_Unity_FC/642 : SUCCESS in 1h 13m 11s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/53/784453/15/check/DellEMC_Unity_iSCSI/7b92845/DellEMC_Unity_iSCSI/591 : ABORTED in 3h 00m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"56bdfb78c0f1a7aff669c64ac0da60b11b93b4af","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-11 20:07:00.000000000","message":"Patch Set 15:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784453-15/PVME-iSCSI : FAILURE in 14:44.70\n* PVME-FC http://os-logs.tristero.net/dell-ci/784453-15/PVME-FC : SUCCESS in 46:20.89","accounts_in_message":[],"_revision_number":15},{"id":"e8634b869859a3c6b9693b26c59f7316efe7275a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 21:43:42.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/784453/15/check/iscsi-driver-nimble/a241a87/a241a87.html : FAILURE in 1h 48m 35s","accounts_in_message":[],"_revision_number":15},{"id":"9636c8d6b128373452a9f98eb6bbf25c3ec0e6e6","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-11 21:50:09.000000000","message":"Patch Set 14:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-14/Seagate-iSCSI : SUCCESS in 42:50.29\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-14/Seagate-FC : SUCCESS in 40:10.84","accounts_in_message":[],"_revision_number":14},{"id":"0e728d7808af8fd449febd381437efce9134e4ca","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-12 04:49:32.000000000","message":"Patch Set 14:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-14/HPMSA-iSCSI : FAILURE in 1:00:12\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-14/HPMSA-FC : SUCCESS in 31:31.82","accounts_in_message":[],"_revision_number":14},{"id":"9b071ba865bd20d131f44cda5f2f842364fd8627","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-12 06:53:25.000000000","message":"Patch Set 15:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784453-15/Seagate-iSCSI : SUCCESS in 41:44.19\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784453-15/Seagate-FC : SUCCESS in 36:55.62","accounts_in_message":[],"_revision_number":15},{"id":"1eb07305cc5b636b870552937e332df7ede11bb7","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-12 15:00:41.000000000","message":"Patch Set 15:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784453-15/HPMSA-iSCSI : FAILURE in 1:00:06\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784453-15/HPMSA-FC : SUCCESS in 36:31.29","accounts_in_message":[],"_revision_number":15},{"id":"67b5d6af0a4bf2b152ec836a8bb962713b94f4ec","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-12 18:24:35.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/53/784453/13/check/dsvm-tempest-TOYOU-ACS5000-driver/ae24092 : SUCCESS in 1h 22m 56s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/53/784453/13/check/fc-test/5542aef : FAILURE in 1h 33m 29s","accounts_in_message":[],"_revision_number":13},{"id":"7eae1d4d17fc036bb0e825dc8360cffc8143b502","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-12 22:58:07.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/13/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"ea501ab5dec13fc2d813bb56f2f820f57a1334cb","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-13 00:19:02.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/53/784453/15/check/dsvm-tempest-TOYOU-ACS5000-driver/21a2dae : SUCCESS in 1h 30m 15s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/53/784453/15/check/fc-test/b1a90d9 : FAILURE in 1h 34m 55s","accounts_in_message":[],"_revision_number":15},{"id":"64d7a261019d97e641cdfdfe996795d8db83136f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-13 04:59:44.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/53/784453/15/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"0e34780273ec9ad5daaf397bcc5c1b2f3e3c13f1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-08-14 00:45:26.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/15/20210814070429/NEC-FC : SUCCESS in 2h 37m 46s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/53/784453/15/20210814070429/NEC-ISCSI : SUCCESS in 2h 40m 53s","accounts_in_message":[],"_revision_number":15},{"id":"2666b6192d475ccfe82b32013df607ba30349dd8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-18 16:33:41.000000000","message":"Patch Set 15: Code-Review+2\n\nRevisions LGTM.","accounts_in_message":[],"_revision_number":15},{"id":"632342c0fb655b61c99eb719b8b4ce137d2ca85b","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2021-08-19 17:13:43.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)\n\nHi Eric Harney,\n  Changes LGTM. I have one query inline, please check it once.\n\nThanks,\nGirish","accounts_in_message":[],"_revision_number":15},{"id":"70d3b1840c34e357c88b83fc9f81b50470f79e3d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-20 14:00:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cf9e2d5d6c612fc1cd142e4a84c2ea11c2ab921e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-20 14:25:13.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"319774624483d999552402b56f649c0fd4f2db56","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-20 14:25:55.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"c898d888612a9cf912048df8e878961573f01621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:26:59.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"0cfee1056bb69fbca64cf356f0650ff70b7adbd3","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2021-08-20 15:06:53.000000000","message":"Patch Set 15: Code-Review+1\n\n(3 comments)\n\nIt\u0027s a nice Change to learn typing.\nIs there a reason you don\u0027t set the return type of all functions, mostly, if there is no return?\n\nDoes mypy also checks types over multiple functions? Or would we need to manually check, if a called function or used object really return this type or use this type as parameter?\n\nI checked mostly the return types and the parameter, if easy to check.","accounts_in_message":[],"_revision_number":15},{"id":"31aec2bd88401f5281f83bc8983f3f3babdcf485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:17:51.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb91f3fdc72e47e0b2e43a51f56aace1 : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f10bd7d59604a62b4493972f1d36749 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e781057e508d423cb8cdfe997d76dfec : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9130b3eef7d4023a7ae56d1959d5689 : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/37db0582a6e34784a883c51d343bda1b : SUCCESS in 1h 09m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a576d7c8bda468c8aed2c601087890f : SUCCESS in 1h 18m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d70f65f7725b40b6b970413221436d8b : SUCCESS in 1h 18m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/858c9ad4259146749b4faf856b2185fd : SUCCESS in 1h 24m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fdb7a9d095646ad85073433bd916100 : SUCCESS in 50m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f717414ad28d47d085c27791722bfdcb : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":15},{"id":"c75afcc9dae1c6a46a2d8268b2257e5ca16ba153","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:17:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"bde8c66e0fda8600dc32abc89b67cfad0ab52e93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:21:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118bf8ef516642e99bc2d96077691bc6 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":15},{"id":"6ea30486a85df569cc5e875853a57b74176caea5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-08-21 10:32:40.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/53/784453/15/ : SUCCESS in 1h 57m 02s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"b5ac2af0c2910b5d8c87a2b6dbd9b6f5f23a7bca","revisions":{"865d455a340056219333deab5e745a93103db587":{"kind":"REWORK","_number":1,"created":"2021-04-01 17:56:30.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/1"}}},"commit":{"parents":[{"commit":"709fdc113860801e6d3da309087480772d817055","subject":"mypy: image_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/709fdc113860801e6d3da309087480772d817055"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:56:06.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/865d455a340056219333deab5e745a93103db587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/865d455a340056219333deab5e745a93103db587"}]},"branch":"refs/heads/master"},"6d135d2f28950846e1f7030436ea650676636856":{"kind":"REWORK","_number":2,"created":"2021-04-01 18:59:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/2"}}},"commit":{"parents":[{"commit":"d581b363b55326879b43a2effcf8efe1d7cacfc0","subject":"mypy: image_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d581b363b55326879b43a2effcf8efe1d7cacfc0"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 18:58:26.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d135d2f28950846e1f7030436ea650676636856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d135d2f28950846e1f7030436ea650676636856"}]},"branch":"refs/heads/master"},"c1234ee8e043b2935dc0ad3a589e8c6cc3698ee7":{"kind":"REWORK","_number":3,"created":"2021-04-27 21:33:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/3"}}},"commit":{"parents":[{"commit":"11fdf27a181aeeb52334f16d73eaa55ef205f33b","subject":"mypy: image_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11fdf27a181aeeb52334f16d73eaa55ef205f33b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 21:32:32.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1234ee8e043b2935dc0ad3a589e8c6cc3698ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1234ee8e043b2935dc0ad3a589e8c6cc3698ee7"}]},"branch":"refs/heads/master"},"67ab00c31ca02712d45fff66bd1f09a642090408":{"kind":"REWORK","_number":4,"created":"2021-04-30 12:26:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/4"}}},"commit":{"parents":[{"commit":"11fdf27a181aeeb52334f16d73eaa55ef205f33b","subject":"mypy: image_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11fdf27a181aeeb52334f16d73eaa55ef205f33b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-30 12:17:58.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67ab00c31ca02712d45fff66bd1f09a642090408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67ab00c31ca02712d45fff66bd1f09a642090408"}]},"branch":"refs/heads/master"},"ff557b7b63340a88196a5da5b70a4cbb2525a44a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-05-03 20:15:53.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/5"}}},"commit":{"parents":[{"commit":"dcfaf2886f269213b1370d10240834f4df025b02","subject":"mypy: image_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcfaf2886f269213b1370d10240834f4df025b02"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-03 20:15:53.000000000","tz":0},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff557b7b63340a88196a5da5b70a4cbb2525a44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff557b7b63340a88196a5da5b70a4cbb2525a44a"}]},"branch":"refs/heads/master","description":"Rebase"},"db880b4a70cfe1945d4252afa055f7470710acce":{"kind":"REWORK","_number":6,"created":"2021-05-10 18:41:26.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/6"}}},"commit":{"parents":[{"commit":"3ae9f172ef5f56bd5924579b94beff665b2d7113","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ae9f172ef5f56bd5924579b94beff665b2d7113"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-10 18:35:24.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db880b4a70cfe1945d4252afa055f7470710acce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db880b4a70cfe1945d4252afa055f7470710acce"}]},"branch":"refs/heads/master"},"ebf840c569dd3b0d465c74da8f1f031e910ab9c1":{"kind":"REWORK","_number":7,"created":"2021-06-30 20:59:37.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/7"}}},"commit":{"parents":[{"commit":"fba2b954cfcfdd99249df3b62ecd8f7b01e54312","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fba2b954cfcfdd99249df3b62ecd8f7b01e54312"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-30 20:57:34.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebf840c569dd3b0d465c74da8f1f031e910ab9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebf840c569dd3b0d465c74da8f1f031e910ab9c1"}]},"branch":"refs/heads/master"},"fa529d09103f0c7367e480fdd2bae17312e1c4ad":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-13 13:49:38.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/8"}}},"commit":{"parents":[{"commit":"16bdba196c8f43215d01b8e5783a80c602d73dd9","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16bdba196c8f43215d01b8e5783a80c602d73dd9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 13:49:38.000000000","tz":0},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa529d09103f0c7367e480fdd2bae17312e1c4ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa529d09103f0c7367e480fdd2bae17312e1c4ad"}]},"branch":"refs/heads/master","description":"Rebase"},"ecf87d4eeebf834721405d2daf05b415e58285e9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-07-15 17:36:36.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/9"}}},"commit":{"parents":[{"commit":"057f06e857e420fb480ec91da3af51cea7f77177","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/057f06e857e420fb480ec91da3af51cea7f77177"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-15 17:36:36.000000000","tz":0},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecf87d4eeebf834721405d2daf05b415e58285e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecf87d4eeebf834721405d2daf05b415e58285e9"}]},"branch":"refs/heads/master","description":"Rebase"},"856242f8c90d7bdcb5a633cbfe2001f026167b7d":{"kind":"REWORK","_number":10,"created":"2021-08-09 13:09:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/10"}}},"commit":{"parents":[{"commit":"1aaaaff35ab3f7faeedcc3180f0f167c8fb53e5a","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aaaaff35ab3f7faeedcc3180f0f167c8fb53e5a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-09 12:56:37.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/856242f8c90d7bdcb5a633cbfe2001f026167b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/856242f8c90d7bdcb5a633cbfe2001f026167b7d"}]},"branch":"refs/heads/master"},"f4eefaa84c2b752db1712d62f9fe43cca29a4db4":{"kind":"REWORK","_number":11,"created":"2021-08-09 15:39:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/11"}}},"commit":{"parents":[{"commit":"111a6251958dd7be8a13ad3a8a9e2a1b711aa09d","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111a6251958dd7be8a13ad3a8a9e2a1b711aa09d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-09 15:31:34.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4eefaa84c2b752db1712d62f9fe43cca29a4db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4eefaa84c2b752db1712d62f9fe43cca29a4db4"}]},"branch":"refs/heads/master"},"f3735f25fd95d0f9ad17794d032238bfd5a27b2f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-08-10 13:24:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/12"}}},"commit":{"parents":[{"commit":"60b07d7e5397d29918abe12e948a8b0de3368dcc","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60b07d7e5397d29918abe12e948a8b0de3368dcc"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-10 13:22:59.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3735f25fd95d0f9ad17794d032238bfd5a27b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3735f25fd95d0f9ad17794d032238bfd5a27b2f"}]},"branch":"refs/heads/master"},"40d0fa4123d69dad3e27886bbbff44a299fcd5d4":{"kind":"REWORK","_number":13,"created":"2021-08-10 14:27:34.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/13"}}},"commit":{"parents":[{"commit":"8c46c09ad5d4ae6b87400b678a2bd7afc4af6631","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c46c09ad5d4ae6b87400b678a2bd7afc4af6631"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-10 14:26:39.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40d0fa4123d69dad3e27886bbbff44a299fcd5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40d0fa4123d69dad3e27886bbbff44a299fcd5d4"}]},"branch":"refs/heads/master"},"9b01a95649a83204eddd865991386ca75f11a88d":{"kind":"REWORK","_number":14,"created":"2021-08-10 14:52:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/14"}}},"commit":{"parents":[{"commit":"8c46c09ad5d4ae6b87400b678a2bd7afc4af6631","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c46c09ad5d4ae6b87400b678a2bd7afc4af6631"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-10 14:51:14.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b01a95649a83204eddd865991386ca75f11a88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b01a95649a83204eddd865991386ca75f11a88d"}]},"branch":"refs/heads/master"},"b5ac2af0c2910b5d8c87a2b6dbd9b6f5f23a7bca":{"kind":"REWORK","_number":15,"created":"2021-08-11 12:41:15.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/53/784453/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/784453/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/784453/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/784453/15"}}},"commit":{"parents":[{"commit":"8c46c09ad5d4ae6b87400b678a2bd7afc4af6631","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c46c09ad5d4ae6b87400b678a2bd7afc4af6631"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-01 17:36:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-11 12:36:09.000000000","tz":-240},"subject":"mypy: continued manager, scheduler, rpcapi","message":"mypy: continued manager, scheduler, rpcapi\n\nChange-Id: I9a8d24ac27af8fe4864934d1b9bc5b66da6d2c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5ac2af0c2910b5d8c87a2b6dbd9b6f5f23a7bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5ac2af0c2910b5d8c87a2b6dbd9b6f5f23a7bca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
