)]}'
{"id":"openstack%2Fcinder~784623","triplet_id":"openstack%2Fcinder~master~I220b5e3b01d115262a8b1dd45758f0531aea0edf","project":"openstack/cinder","branch":"master","topic":"rbd-encrypt-issues","hashtags":[],"change_id":"I220b5e3b01d115262a8b1dd45758f0531aea0edf","subject":"[rbd] Fix create encrypted volume from snapshot","status":"MERGED","created":"2021-04-03 04:38:13.000000000","updated":"2021-08-10 02:07:02.000000000","submitted":"2021-07-30 19:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":24,"has_review_started":true,"submission_id":"784623-rbd-encrypt-issues","meta_rev_id":"6118c65bf34565cc86241886de2ffce3dde31345","_number":784623,"virtual_id_number":784623,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-30 19:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2021-07-29 18:31:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2021-07-30 17:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-07-29 17:13:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2021-07-30 17:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-03 05:05:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 05:38:56.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-03 05:54:40.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 06:04:14.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-03 06:07:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 06:21:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 06:22:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 06:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-03 07:29:20.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-04-03 09:11:39.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-03 09:18:05.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 09:19:33.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 09:23:43.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 09:27:52.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-03 13:01:36.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 13:19:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-03 13:41:28.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-03 19:44:30.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-03 22:12:37.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-04 05:03:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-06 12:41:08.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-06 13:20:06.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-06 13:50:52.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-04-06 16:04:27.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-04-07 02:12:42.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-07 07:00:44.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-04-09 01:37:22.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-04-10 19:40:48.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 02:27:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-22 16:01:49.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 05:03:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-03 13:17:43.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-05-08 06:49:19.000000000","updated_by":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"reviewer":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"state":"REVIEWER"},{"updated":"2021-05-10 06:10:37.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-05-17 17:25:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-05-18 15:08:29.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-07-27 19:31:50.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2021-07-28 14:46:39.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-08-03 04:13:59.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"}],"messages":[{"id":"894d377ead14821ce0052e70a321d0231c1e1871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-03 04:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ef535079a26bfa1f89d7cc7fe78feff53cc47e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-03 04:39:24.000000000","message":"Topic set to bug/1922408","accounts_in_message":[],"_revision_number":1},{"id":"8115f16508e8ede7f0bc52dcb629205ad8443404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-03 04:39:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"84e0871bf7f5b238e9e2829fe9748c8d6aed52f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 05:05:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea5e8e78d14b4d53ac61dcfe58d98d18 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5984a7e42f54c069b0833bf152b8cff : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b44db4fcfc769d43a732882c9926c8ccd1d6cd07","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-03 05:38:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/2/iscsi/ : SUCCESS in 58m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/2/smb/ : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"6a25a689886a246ef340db7fe8cb669b79161ae9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-03 05:54:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/2/check/dsvm-tempest-zadara-driver/6ac7c38 : SUCCESS in 1h 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0425d6b1b11739e95cda747391b9f97e0647fa2e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-03 06:04:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/2/check/dsvm-tempest-cisco-zonemanager-job/68f6bf0/ : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f020eac133df63fcc56ae288cdd30294d7e20684","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-03 06:07:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/2/check/iscsi-driver-nimble/66aad92/66aad92.html : FAILURE in 1h 23m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9d63df64a43954b8e705614c9c25baf9490f5fe9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-03 06:13:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/61f06f4/61f06f4.html : SUCCESS in 1h 32m 25s","accounts_in_message":[],"_revision_number":2},{"id":"91484be3129da0aefe88d7ab62230291cdebd839","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-03 06:15:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/2/check/iscsi-driver-primera/c66f2e7/c66f2e7.html : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":2},{"id":"edba5ef56c6f189c539f2ffdeae0b374af1ac334","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-03 06:21:49.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/2/check/tempest-dsvm-full-focal-py3/d28b782/ : SUCCESS in 1h 37m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/2/check/devstack-plugin-ceph-dsvm-focal-py3/51c1498/ : SUCCESS in 1h 32m 45s","accounts_in_message":[],"_revision_number":2},{"id":"03dec3ea6090fb5ddece815db1d7fd757757e349","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 06:22:41.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"4c41b7f79096d74fc0f3e0fa4c8f4fab5c0d2fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 06:29:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751bb29b35ef4d3d9771516462f98516 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c5789a6e884ab08d3b3bd1ad8700cb : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/460a3d06bfd240199c291a042a07d69c : SUCCESS in 11m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b5dcd98db64ab18ef4080f1683817f : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73b608ec275c4cfa87cec5663d5ea11f : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b8b650c6104e118ff7e067a02d8cd3 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f896e10b83544b059757ba14c65223ff : SUCCESS in 1h 06m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ccb8baad14647509afb012e74d7ef59 : SUCCESS in 1h 21m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb18ac2e38ec4ea0b7f7151ea4fbae25 : SUCCESS in 6m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f4fa3c2c8ef4d2d98c0b946180d2018 : SUCCESS in 12m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/25eac877274c4cd48375b6615f89f688 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bad3ffeb604149ccafaa2136e0348139 : SUCCESS in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb62a2f5ccc44e73b560207f362107e9 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5caf0d575ce48d497028c444013774b : SUCCESS in 40m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2fc13140d964d5885fd2ba0cf22cfe2 : FAILURE in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/568920f54a77440a9e047cf69b0937b1 : SUCCESS in 1h 12m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/baccb93932324b71bd62c1bc230c8183 : FAILURE in 1h 41m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dfaaa6defe494c2ea5c16bed601e1441 : SUCCESS in 1h 12m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c1bcdf2e1914007b98c80002d3f7c3d : SUCCESS in 1h 20m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3cc238f994f943ef9d7940a83c8673a5 : SUCCESS in 1h 17m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6658790677ea4eba92417993b4d2c194 : SUCCESS in 1h 03m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8116ac55bc6c4a3d9ad42751287b6e64 : SUCCESS in 1h 48m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/496fbbb496344f45986b914aa5ce7f6d : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":2},{"id":"346b8c7af39f8341f5ac72de06fbe62b423f678b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-03 07:29:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/2/check/DellEMC_PowerStore_iSCSI/a5b7121/DellEMC_PowerStore_iSCSI/736 : SUCCESS in 59m 06s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/2/check/DellEMC_PowerStore_FC/f7959e6/DellEMC_PowerStore_FC/731 : SUCCESS in 54m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d499edbd54b30549eed415176102602c86081d0c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 09:04:32.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fb717502a8f0a23d07a6fbc18d1443cced456af6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-03 09:11:39.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/2/check/DellEMC_SC_FC/f3b3731/DellEMC_SC_FC/578 : SUCCESS in 2h 17m 20s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/2/check/DellEMC_SC_iSCSI/12f22bf/DellEMC_SC_iSCSI/575 : SUCCESS in 2h 41m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"3c462b100aeba199abc44074aaf021767e4f9e96","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 09:18:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/2/check/DellEMC_VNX_iSCSI/be3d59f/DellEMC_VNX_iSCSI/706 : FAILURE in 1h 13m 51s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/2/check/DellEMC_VNX_FC/aec82a0/DellEMC_VNX_FC/755 : FAILURE in 54m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"000e56a718f96903d61747a11c3dc6ed00fc534e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-03 09:19:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-1/ns5_nfs : SUCCESS in 4h 40m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-1/ns5_iscsi : SUCCESS in 4h 34m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"df42fa32dabc8d3a28ed6361d7a953a64054a1c2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-03 09:23:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/2/upstream-check/cinder-cDOT-iSCSI/3b51694/ : SUCCESS in 41m 06s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/2/upstream-check/cinder-cDOT-NFS/6f5316f/ : SUCCESS in 42m 04s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/2/upstream-check/cinder-cDOT-FCP/e09654c/ : SUCCESS in 1h 01m 14s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"68fa0256dec0ddb34ea68099c5fadb58d976e85e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-03 09:27:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/2/upstream-solidfire-check/cinder-solidfire-iSCSI/00e3f2f/ : FAILURE in 55m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"1217ee8bc500883542b0b13de160371a2bd540d7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-03 10:06:31.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-2/ns5_nfs : SUCCESS in 4h 42m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-2/ns5_iscsi : SUCCESS in 4h 33m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5d85b5931a7c4e6edfcbad0078192efed017596a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 11:38:30.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"84111db24409ee3104c80bd8883e7ec483c55f43","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 13:01:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/2/check/DellEMC_XtremIO_FC/174b583/DellEMC_XtremIO_FC/1230 : FAILURE in 2h 03m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/2/check/DellEMC_XtremIO_iSCSI/7b29bed/DellEMC_XtremIO_iSCSI/1250 : FAILURE in 3m 25s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"88fdbc1170103c5dcef9c28cb2e308dd8497093a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-03 13:19:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/2/check/DellEMC_Unity_FC/0055aac/DellEMC_Unity_FC/1079 : FAILURE in 4m 34s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/2/check/DellEMC_Unity_iSCSI/8f32efa/DellEMC_Unity_iSCSI/1016 : FAILURE in 2h 01m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fc1c113fbc3bf1360f70fddf6bc6f8a073e23177","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-03 13:41:28.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-2/PVME-iSCSI : SUCCESS in 56:51.95\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-2/PVME-FC : SUCCESS in 55:35.08","accounts_in_message":[],"_revision_number":2},{"id":"89f11b7158041c9bf7a2b456983dff12d1d114ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 15:38:03.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3969a8b94fca5d4ac94a2402792958bf58b32f67","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-03 19:44:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/2/check/DellEMC_PowerFlex/b5ffc0c/DellEMC_PowerFlex/727 : FAILURE in 2h 46m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3410cb6f30b520434e10462b55f939c3ef6e168","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-03 20:58:59.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a39f763b085098054bea31b4ef82c02cfaaa2597","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-04-03 22:12:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/784623/2/ : FAILURE in 30m 25s","accounts_in_message":[],"_revision_number":2},{"id":"9809d6a2ef9aa345c5b3b818a90cbfadfe415fb3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-04 05:03:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/2/20210404113817/NEC-FC : SUCCESS in 2h 21m 25s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/2/20210404113817/NEC-ISCSI : SUCCESS in 2h 25m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6d6a8eada066b33a1b4aa8ab1e99c99f587d6818","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-06 12:41:08.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-2/Lenovo-iSCSI : SUCCESS in 48:40.97\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-2/Lenovo-FC : SUCCESS in 44:14.23","accounts_in_message":[],"_revision_number":2},{"id":"9c04261dd66f2af461345bc8361b4be3117c6068","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-06 13:20:06.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-2/Seagate-iSCSI : SUCCESS in 42:54.62\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-2/Seagate-FC : SUCCESS in 38:19.96","accounts_in_message":[],"_revision_number":2},{"id":"4b8b2b4abe39d7cf92d63557ea0bf11710c1471b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-06 16:04:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nHi Haixin, thanks for working on this. @eric and i been looking and i leave some comments in line:","accounts_in_message":[],"_revision_number":2},{"id":"7d036f8fbf0df760d03b138ac86e6014ca56548b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-07 02:12:42.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-2/HPMSA-iSCSI : FAILURE in 59:23.83\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-2/HPMSA-FC : SUCCESS in 33:14.41","accounts_in_message":[],"_revision_number":2},{"id":"98464ec7d86bc691ff8fb93489a45009d89a110e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-07 06:35:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7dbe40aefe266887f660e8b877780262f2fcec62","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-07 06:36:37.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nhi, Sofia Enriquez.\nthanks for your review.","accounts_in_message":[],"_revision_number":3},{"id":"9e4ae4b9a7755097e0ff92bd42b0119c0cbce4f2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 06:39:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 1m 22s","accounts_in_message":[],"_revision_number":3},{"id":"1e2b5bd53d1b188134a5a6c544a6897f1d3370f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 06:44:29.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1d321c168584c9e70be1193ef4d781cf9936ae56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 07:00:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63eb023d755442efbebf3670d24cbfdd : SUCCESS in 19m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7b205937cc6401a85f7e59841a3cc30 : SUCCESS in 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cebc0f4c5df8346965b86f508f99544d835ba767","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-07 07:14:48.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-3/ns5_nfs : FAILURE in 38m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-3/ns5_iscsi : FAILURE in 27m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"2d144651ab0f03cb1cd12a4df030e309d4ff6d5f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 07:22:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b8698067a0088b22773c8829dff7827d0306cbba","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 07:35:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/3/iscsi/ : SUCCESS in 59m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/3/smb/ : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b98d980ce46959b4746c8c0909dae55ce4c3b402","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-07 08:10:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/3/check/iscsi-driver-primera/d5c98c1/d5c98c1.html : SUCCESS in 1h 33m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b558e0bfbee63b968466ca723fb9f9ac9c8d32f8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-07 08:15:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/3/check/dsvm-tempest-zadara-driver/6027e6f : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aba70274024dd2bc6e2a3bd3e88bf7d781458089","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-07 08:18:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/3/check/iscsi-driver-nimble/8bf4211/8bf4211.html : FAILURE in 1h 27m 07s","accounts_in_message":[],"_revision_number":3},{"id":"4a42b8960a2f75a8d341280fa30ed86f37013daa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 08:22:13.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a7d6a79942988834dff111866ebff7791df08a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 08:35:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0044e52541742fc91708da5d872c5df : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540f9b5325934249b75f4975a507f870 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95237537dbc84bb39cb35cb1c4723fe8 : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a439c6dd8394b8bae62306b5241f7d4 : SUCCESS in 18m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480f2c4cf3b34a958e9e36a5dee41202 : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f3bf194c3a4aca99adc18fc442b812 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0ba3e8e56b4d44119cd97ebb904397d8 : FAILURE in 15m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed27a1134c1749debdba5787ce2fc0fc : SUCCESS in 1h 01m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af631c102d2b44ef815dcc1a1236e8a8 : SUCCESS in 11m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aaf983976cfa407c915fe05ef18f09ed : SUCCESS in 14m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/25fb355afa444b2e8a6dbe47d52bd85e : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e95e8ba499f44617afdb92337e54bfb4 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8c8b0715b7e44cdbc8f80439e6fced3 : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1fc9da68cd9c43f18da0c09bd67e7e8e : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/658f78c9371743b0bf97226e5013d32c : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9dcc38512db46b7a887e31092d8a8c5 : SUCCESS in 1h 36m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cad04151abf043a1b58a835129708a89 : FAILURE in 1h 32m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/82fd10c09d484e93a35e5d3955972684 : SUCCESS in 1h 38m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ef5cb1d560a4663982c8cdcc62798e4 : SUCCESS in 1h 20m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85e347a4ae9b483fab0ea75e64f0b13b : SUCCESS in 1h 23m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc7bb296495044508a1f10189cb0e537 : SUCCESS in 1h 23m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52319b75b13e4d36bc80b4d29e1aaf7d : SUCCESS in 1h 49m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a821bd61998441a48b44510acf9a0d04 : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":3},{"id":"01c57e9e793173c9f5615e7a5248016ba1c03b94","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-07 08:47:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce8efafc9d860a4a9951819fa97bfdce7a760843","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 08:51:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 1m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0c7af520ae66d247b3a148a316b55e9f93981251","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 08:55:23.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f14bdf12e147f37f9766af88e324623e62486b20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:18:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f69bd1d9e3ac472a9a3dc3c58284d6db : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27bd2fc89c7e4b2ca0b818a920bff0e7 : SUCCESS in 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a25412572ee3c32432dab5b9963ae10b8b44f478","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 10:03:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/3/iscsi/ : SUCCESS in 1h 15m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/3/smb/ : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f4865a3b5a791593f79d699a60c074ab72c3b111","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-07 10:03:27.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/3/check/tempest-dsvm-full-focal-py3/f21af28/ : TIMED_OUT in 3h 08m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/3/check/devstack-plugin-ceph-dsvm-focal-py3/26fb55f/ : FAILURE in 2h 11m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ae91f6c17d52e7b7d8d82fff866c66db327ed5b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 10:11:47.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"ab28c13deffe40a8f9e5d29e6de10d7a7e382834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 10:50:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40122e8e1fb642c5934c07e908120dba : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf77b15faee49068b8fd67c9d43057e : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8dec8a09d974d6397b13e35af58bfa8 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a52b0915c79949b2b19306cc98e5c19c : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/171ebce21b8c42aa805c17de58ffcf38 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1df36f7b08849a48565c5f7d3f6b07a : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/ddd9cecaf66a40a8b823ec89e3a45c6e : SUCCESS in 1h 07m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/448c6bb8d63d4649a95a39eff8d1ffd7 : SUCCESS in 1h 27m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b92cc4823c46eba87c1f7dabcf12e0 : SUCCESS in 6m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8922a706979b424c950133403861db3b : FAILURE in 15m 59s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf29b548d59d4f24807060bf9644f547 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53dde3c67b30423e91c664f95273a758 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b46ef48d39a4eb699bd0e7d3d0203f2 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/786fb3db452a462c8524a0c2a4d02396 : SUCCESS in 48m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40a6f26feafd493bbdc397430087b209 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48cb47e85c1047d9911cc1f4e8ffa0ac : FAILURE in 1h 28m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cc7247f632d843488a7d43171dd7e23b : FAILURE in 1h 33m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/301259f27078415b901c7f0fec6df0ad : SUCCESS in 1h 43m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/28debc9a58f8426998fa691aab682a2f : SUCCESS in 1h 34m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6e2d6788405d4f758e42107599215cdb : SUCCESS in 1h 20m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/399f53f7f26741aab995be47e23ecda2 : SUCCESS in 1h 38m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67cda21bc5144da298c1c5b6d5a4ff8c : SUCCESS in 2h 01m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1e203a748b14cda99eba542133d1966 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":3},{"id":"3d988b8f93f5c19e8b951351a2d5a214d8607e55","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-07 12:22:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/3/check/dsvm-tempest-zadara-driver/d37c772 : SUCCESS in 1h 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b00ceb410eae4e9c3915a4282a10a171df811196","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-07 13:32:55.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-3/ns5_nfs : FAILURE in 32m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-3/ns5_iscsi : FAILURE in 32m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a5892fecbbec121581a5dfc1c19366b5ca6938df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 13:42:17.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"ec2fedd24fca78c65cd127b9324183f770238fc5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-07 14:09:19.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/3/20210407222211/NEC-ISCSI : FAILURE in 43m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/3/20210407222211/NEC-FC : FAILURE in 47m 03s","accounts_in_message":[],"_revision_number":3},{"id":"a0bfbf0854f007cec05a9de3c91cf050848741ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 14:22:14.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"10da1c08e03b8b1dfae29271502893d64c59805a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-07 14:58:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/3/check/DellEMC_XtremIO_FC/8aee04f/DellEMC_XtremIO_FC/1254 : SUCCESS in 2h 06m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/3/check/DellEMC_XtremIO_iSCSI/076683e/DellEMC_XtremIO_iSCSI/1274 : SUCCESS in 1h 47m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"720ade9a8d4c8f55935426f549740da556dea3f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 15:08:13.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"01a9f4c6c412261a56f1c34edc6d0e216ee63d5f","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-07 17:48:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4d6a0e52d9b616777d2fbe1b96b09d40b7e44483","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 17:57:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"45264afe5903347f6e0ddf8e0cd1eedb4748a98f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:10:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081c0dbce2824cc6bccba1f9e44c7539 : SUCCESS in 16m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12669626b2e049dd840e5d00c0e5eaf0 : SUCCESS in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb73db061ce2b4434d267333ead1acc4bb5363b2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-07 19:05:06.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-3/ns5_nfs : SUCCESS in 5h 25m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-3/ns5_iscsi : SUCCESS in 5h 20m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"e0602f713d0997ad1ef615998718076ccd928323","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-07 19:08:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/3/iscsi/ : SUCCESS in 1h 07m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/3/smb/ : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8292cba2a538d1413efea6ee0e9bdbfe115c6687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 19:34:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/809b6f1ca8bc4fe2961865b67f9a633b : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77d3f533f5f414da5b63630a3ee0fe6 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ac957df2a2a40cca20d13c8d5e28e6c : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb05295565844dffaeffa5fe25fd1a43 : FAILURE in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/157f957f0f784d6ea8a20c907e39c54f : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab54ccae9a0c4dceb48bd6c623d162ab : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/73e81aa2c0fb4d47a1ceda353548ae9c : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/684d9471feee40b0abdbe40f1ffe05ac : SUCCESS in 51m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0ed3b08dcc549d3926691ee1d5251e9 : SUCCESS in 7m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90f415a41f164785876eb14a04f0eb7d : SUCCESS in 12m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4999a6a7124443199300f94ba5663139 : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13ef475a1d81415d87c3cc00d9216f46 : SUCCESS in 6m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa123768bd814c4daf7b7cf0893de397 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6f075584e1e4898859700589e71846e : SUCCESS in 47m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38a61ac2d6e84011af6f47c9ec6f1e19 : FAILURE in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19be593576e34001883566cf4b001c0a : SUCCESS in 1h 35m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/87051fecfeed4464bc1693c9fcdecbee : FAILURE in 1h 29m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6980faf0450746a18435f83a58a4deb4 : FAILURE in 1h 39m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a3441c105f68404792c1a481b229df6c : SUCCESS in 56m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/658701d7d7994f46a2c476e8a82e268f : SUCCESS in 1h 29m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fffe52041d514b46a11c04959e9816ca : SUCCESS in 1h 34m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36d75a0a57484b2584650c9c3b854466 : SUCCESS in 1h 38m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1172e177908344c1b82b5125b27320ce : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":3},{"id":"4d8209f6ad09317be88d88fa41ea57677d731dcb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-07 19:40:07.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/18f95be1e9d043e0b1f20c50c19acff6 : RETRY_LIMIT in 46s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b5dbd0f31f8849a2825f03ae00428862 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":3},{"id":"25ea14427cc7fe9ba96d009f9f46304d957e8006","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 19:40:58.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"1544f10024602b89d8c1853caaa5dfda2b345fc3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-07 20:12:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/3/check/dsvm-tempest-cisco-zonemanager-job/cb817e4/ : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":3},{"id":"dc3719bcf19c5b99b6aa8a9c1f5bf66dac8f6f5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 22:17:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c9700e510ba21a64369a3d4734c144cb919020e8","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-08 00:47:17.000000000","message":"Patch Set 3:\n\nthe error is ERROR:   full: could not install deps.\nmaybe some wrong with zuul.","accounts_in_message":[],"_revision_number":3},{"id":"6250943d99814e5b30b8d0984fc1082cd6f0a9ef","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-08 13:29:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4cdfa8d5217f2ff056a22946d891e04e39e3a9c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 14:27:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78f6476296d441b3bc2b04d6fdb24594 : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf94133159e3463a9327ef7b45c07e33 : SUCCESS in 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21b0181ce5374508f7ddca3eaf907fa5dbe1cd14","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-08 14:44:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/3/iscsi/ : SUCCESS in 1h 14m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/3/smb/ : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":3},{"id":"40545b3cded2457bdeb25e0855379d02b1c2856e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-08 14:51:53.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"69f37b1cc24583af7b447b4835e25c6ef2a1891a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-08 14:55:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/3/check/dsvm-tempest-cisco-zonemanager-job/96113c9/ : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2bd7e71a38ca3d11234e45458d868b935dc7a7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 15:57:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c06632ec043e4a60b46194f449ec66fd : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8d77903f7542e398571b09197143ef : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c649c812b9a3444aa50b87d9ce77f484 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afd3e06383a34e7eb98710fe3b9f3bbe : SUCCESS in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf11abecaa634eeb84bdd8eaca754833 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09795c3c30554a7396e14cd7a6cedfda : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/974f4aa971ba4d4d8fb028421c070ab1 : SUCCESS in 51m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abd3fe506d0045c8a931a9ffc5fb4cf3 : SUCCESS in 1h 18m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7ad221b27254a7589e9b46498b2f5a2 : SUCCESS in 7m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9cf5d3e28c54f5b940744f5e657cca5 : SUCCESS in 15m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/26352b31501a4ba2870df44b06897c82 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7642c422de0a40f0894ae2cab3b390bd : SUCCESS in 7m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54561470baa445c18fe10d2e5fbc57cc : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b16baa19911a46c1861b9f01dc08539c : FAILURE in 35m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2912e9282d245d0b1c6e78a4b7acd6f : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bda4f802757147c0ae3eecd34313aef1 : SUCCESS in 1h 16m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aad2c1ded33e4eb881aae87b92406dc8 : FAILURE in 1h 35m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/168d9161634647798432d8f316c7affe : SUCCESS in 1h 34m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e757bf8f81e94e98816637a6b90ed50b : SUCCESS in 1h 03m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c309181518ad4f5a8a2662130af0946a : SUCCESS in 1h 20m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1caddedf44de4ab684f15192fa79f714 : SUCCESS in 1h 31m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fb645ce53e948d7b3d92671d8cff912 : SUCCESS in 2h 05m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67b5007ff1644e19803bcaabcf541d87 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":3},{"id":"34b7181674d9a7ac658423295ccefc3567c573b7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-08 16:05:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/3/upstream-solidfire-check/cinder-solidfire-iSCSI/be1522f/ : FAILURE in 5m 14s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"ee269ae2a3241ac10a4bf73b915b7e66191f15ac","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-08 16:22:46.000000000","message":"Patch Set 3: Code-Review-1\n\n:-1: Please add unittest","accounts_in_message":[],"_revision_number":3},{"id":"b8f24e535a740d1dce23c6d2fa45a6a65ea1a763","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-08 16:42:01.000000000","message":"Patch Set 3:\n\nApart from the unittest, looks like this patch fixes the problem:\n\n```\nenriquetaso@enriquetaso:~/devstack$ sudo rbd info volumes/volume-863b595a-8bd4-4cbd-8e39-2cae3bbf79fe\nrbd image \u0027volume-863b595a-8bd4-4cbd-8e39-2cae3bbf79fe\u0027:\n        size 1.0 GiB in 257 objects\n        order 22 (4 MiB objects)\n        snapshot_count: 1\n        id: 15d820ba1574\n        block_name_prefix: rbd_data.15d820ba1574\n        format: 2\n        features: layering, exclusive-lock, object-map, fast-diff\n        op_features: \n        flags: \n        create_timestamp: Thu Apr  8 16:35:10 2021\n        access_timestamp: Thu Apr  8 16:35:10 2021\n        modify_timestamp: Thu Apr  8 16:35:10 2021\nenriquetaso@enriquetaso:~/devstack$ sudo rbd info volumes/volume-b4291cbe-ef15-45f3-8234-69ab47860758\nrbd image \u0027volume-b4291cbe-ef15-45f3-8234-69ab47860758\u0027:\n        size 1.0 GiB in 257 objects\n        order 22 (4 MiB objects)\n        snapshot_count: 0\n        id: 1608db36a9d8\n        block_name_prefix: rbd_data.1608db36a9d8\n        format: 2\n        features: layering, exclusive-lock, object-map, fast-diff\n        op_features: \n        flags: \n        create_timestamp: Thu Apr  8 16:38:18 2021\n        access_timestamp: Thu Apr  8 16:38:18 2021\n        modify_timestamp: Thu Apr  8 16:38:18 2021\n        parent: volumes/volume-863b595a-8bd4-4cbd-8e39-2cae3bbf79fe@snapshot-29236f30-a609-41d0-8855-d63ec0898afa\n        overlap: 1.0 GiB\n```","accounts_in_message":[],"_revision_number":3},{"id":"f27373768899b70b2e1a9883e31c6ae43b56c93a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-08 17:02:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/3/upstream-check/cinder-cDOT-iSCSI/094f30f/ : SUCCESS in 38m 57s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/3/upstream-check/cinder-cDOT-NFS/645b277/ : SUCCESS in 41m 56s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/3/upstream-check/cinder-cDOT-FCP/2ffd4a7/ : SUCCESS in 1h 02m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2a153ca64fdbae7403f78c74bafa37362e689aef","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-08 17:03:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/3/check/DellEMC_PowerStore_iSCSI/45cb70c/DellEMC_PowerStore_iSCSI/758 : SUCCESS in 1h 04m 50s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/3/check/DellEMC_PowerStore_FC/e241dbc/DellEMC_PowerStore_FC/753 : SUCCESS in 55m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"404a4386d1b4ee72a8b17dc43d5bc290aed0d366","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-08 17:51:45.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/3/check/DellEMC_Unity_FC/3ffab14/DellEMC_Unity_FC/1094 : SUCCESS in 1h 32m 34s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/3/check/DellEMC_Unity_iSCSI/2b5e36a/DellEMC_Unity_iSCSI/1031 : SUCCESS in 1h 53m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"df234c890115ebde824d702737e0bc33523d55cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-08 17:52:42.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"e3f0c91ce4bc23d513525cc811f8cf03aca2d1be","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-08 18:01:47.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-3/Seagate-iSCSI : SUCCESS in 46:32.95\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-3/Seagate-FC : SUCCESS in 39:06.68","accounts_in_message":[],"_revision_number":3},{"id":"2154206e5a2caf1b48e12e9675bb7603163c95f6","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-08 18:06:15.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-3/Lenovo-iSCSI : SUCCESS in 47:59.87\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-3/Lenovo-FC : SUCCESS in 43:45.55","accounts_in_message":[],"_revision_number":3},{"id":"5fa99a32e67bfd27d37bfa282608bb41f64faa56","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-08 18:06:49.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/3/check/DellEMC_SC_FC/4d4d94f/DellEMC_SC_FC/600 : SUCCESS in 2h 05m 34s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/3/check/DellEMC_SC_iSCSI/b9aab5b/DellEMC_SC_iSCSI/597 : SUCCESS in 2h 08m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"c4138c796b255640c9850475c5648f60eecdd3f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-08 18:07:17.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"b619c89bf4ac0e354fde3a36ed63a03875675650","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-08 18:09:00.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/3/check/DellEMC_XtremIO_FC/f761277/DellEMC_XtremIO_FC/1267 : SUCCESS in 2h 04m 18s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/3/check/DellEMC_XtremIO_iSCSI/802ddfa/DellEMC_XtremIO_iSCSI/1288 : SUCCESS in 1h 47m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"3431f87984b45719022f3fab64c3a86be4ba7076","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-08 18:09:18.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"c0e4b945aede21df9346baed39e1df58e889bff1","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-08 18:22:53.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-3/PVME-iSCSI : SUCCESS in 59:23.03\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-3/PVME-FC : SUCCESS in 51:27.11","accounts_in_message":[],"_revision_number":3},{"id":"e981fb5ac79bfb904ae711cef1d8f2fcb7509545","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-08 19:02:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"701ec16d8518267ef7b785dcb376273aca2da3d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-08 19:17:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/23/784623/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"fcefcf43f5fd84a83b41b19374c83493edf7e367","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 19:25:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae18ff041b424b9dbb1054a29c0baf81 : FAILURE in 18m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/833b907df3b24209bbbb24e7bfdb6715 : FAILURE in 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"665f1cbcd8301c60dcc8fd84d0af43cd892c3169","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-08 20:09:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/4/iscsi/ : SUCCESS in 1h 06m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/4/smb/ : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":4},{"id":"399c0b251234f06ea4d8225be94b3983e79e526b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-08 20:16:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/4/check/dsvm-tempest-cisco-zonemanager-job/d2275b3/ : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":4},{"id":"2e829df31d18654d98e672c577fb028f15d0131b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-08 20:30:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/4/check/iscsi-driver-primera/f8c643d/f8c643d.html : SUCCESS in 1h 27m 20s","accounts_in_message":[],"_revision_number":4},{"id":"16aa8b6b749d9e68427a343e7cacca653cf8535f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 20:53:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b5fc49d99244bdf96905b333434173c : FAILURE in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b1dd4ff7a54d4e8056024259d6c0d8 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9abf030439724bdf833b39e9b7ef234a : FAILURE in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c02a03a3ae1b41c59175fc4a6ca6e8a5 : FAILURE in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab47efbe9e4d4f2796734630ab7fffc6 : FAILURE in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ede63d8f2f47ae8417f9a201e4d027 : SUCCESS in 14m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ff6b1f12402b4589b6d9e60096b52ce6 : SUCCESS in 52m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/499f2dee4f844f9b85216ba82bf09942 : SUCCESS in 1h 20m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/284a4840645340c182bd5798bc2bdf92 : SUCCESS in 8m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f4f873e21c24893b28b8a2356f09990 : FAILURE in 13m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de74a19a33304f348acdce7f77388e52 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98f48a2b0d654797819e81b3969e6cea : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5fca9c5036e411cbf43f38acaab0ca8 : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/39d9344994d84ae4a47fc100ddd359d8 : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88ac43bebe784129b34d64bfc483b3a2 : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/843366aa8d9e44cf904ce1ba5d747f0e : SUCCESS in 1h 07m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/69e6be36117a419d9b128614df5bf4a2 : FAILURE in 1h 40m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/098cd43e80fc4ca6936cea493b0ab831 : SUCCESS in 1h 24m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b94db26c30ec471ca2163047badab157 : SUCCESS in 1h 15m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93ef2864f6d24bb3a740a47de25b0fab : SUCCESS in 1h 17m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77f8827d70884489ba78cd4f7bfcbedb : SUCCESS in 1h 14m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ea155af13744d3cab6968f40f7db64d : SUCCESS in 1h 49m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/189bf11c5e94441483c21e029aab19da : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":4},{"id":"155d138b8a2bfbb0a62f27a510e385cbf30ddadd","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-08 22:18:40.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-3/HPMSA-iSCSI : FAILURE in 56:57.77\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-3/HPMSA-FC : SUCCESS in 33:30.80","accounts_in_message":[],"_revision_number":3},{"id":"c31723c430d912046a06eb293141dd3cae8be0da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-08 22:31:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/4/check/iscsi-driver-nimble/b797fc4/b797fc4.html : FAILURE in 1h 10m 06s","accounts_in_message":[],"_revision_number":4},{"id":"844d3215bf4c630bbd0223b82676041e1c45142d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-08 23:20:50.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/4/check/DellEMC_Unity_FC/abba05e/DellEMC_Unity_FC/1097 : SUCCESS in 1h 24m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/4/check/DellEMC_Unity_iSCSI/b4e09ad/DellEMC_Unity_iSCSI/1034 : SUCCESS in 1h 43m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"54b16777a4721eb8e2c540ba72c2436f5682f1c6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 00:06:59.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/4/check/DellEMC_XtremIO_FC/2aa6f29/DellEMC_XtremIO_FC/1270 : SUCCESS in 1h 58m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/4/check/DellEMC_XtremIO_iSCSI/86c60e9/DellEMC_XtremIO_iSCSI/1291 : SUCCESS in 1h 43m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"d9aef64dd503db2859556de6ae9585831f206017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-09 00:18:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba81b603eed0801a0937ad0b564119998bc96955","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 00:41:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/671489522f1d4a1d92e4bcaef8be97d5 : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b70cab6ee53c43c689dc54806fec386a : SUCCESS in 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"25ffc878e40085207a628b48f55dbd8d8d14cb2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 00:49:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/5/check/iscsi-driver-nimble/a41e41c/a41e41c.html : FAILURE in 29m 41s","accounts_in_message":[],"_revision_number":5},{"id":"2c5460e937ea08bd937a9f3b29ad0a7f3930c0a2","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-09 00:50:37.000000000","message":"Patch Set 5: Code-Review+1\n\nthanks for add unit test.","accounts_in_message":[],"_revision_number":5},{"id":"921def01d7b591a79ecd67dc176177d31878828f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-09 00:51:45.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-4/ns5_nfs : SUCCESS in 4h 52m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-4/ns5_iscsi : SUCCESS in 4h 43m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"3c2f640d930a15b14f13e99a2def02c7cd3b7aaf","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-09 01:04:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"648a4110820efada2cde72ad9ff0062bd43ff3b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-09 01:09:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"62eb3419f628519b181294d03fb86746da6d352e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 01:29:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e81f8ed589374df4905fcd29c0ceecf0 : SUCCESS in 15m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67238ad422c84dcc811384475b1224c4 : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"220992e0bb46aa2bd8288a5cba5c7b32caa32050","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 01:37:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/6/check/iscsi-driver-primera/464213e/464213e.html : FAILURE in 26m 25s","accounts_in_message":[],"_revision_number":6},{"id":"48c9cd8613743dac972276aa3381d0d83fb97a9a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-09 02:07:35.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/6/iscsi/ : FAILURE in 57m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/6/smb/ : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":6},{"id":"69aa1cd739d0640b5e4af8dbc9a5345b0dfce717","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 02:22:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/6/check/iscsi-driver-nimble/c8c8ae3/c8c8ae3.html : FAILURE in 1h 11m 50s","accounts_in_message":[],"_revision_number":6},{"id":"0dc6ba1db27aaf1db9547f89d8117881748afd2f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-09 02:31:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/6/check/dsvm-tempest-cisco-zonemanager-job/e7adf8e/ : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":6},{"id":"174478a670aa6837eab1a87caa5c6a207eac1093","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-09 02:44:56.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_Unity_FC/f9c31c4/DellEMC_Unity_FC/1100 : SUCCESS in 1h 14m 43s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_Unity_iSCSI/a611dec/DellEMC_Unity_iSCSI/1037 : SUCCESS in 1h 34m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"a192bc4e761e01bf62a0bca8156f5b8d83c6b2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 03:26:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9119495156a4468a81de4d9281af0d4e : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060feb3ef4e048ac842e1943fb62def3 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac8099c69be64c23b6a620252aa8e862 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84d6feae32c341b4872400d7f41d9b68 : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a52fcee0fbc4112acba96b1cc5db4f7 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/890641bd30f9402bbc606b747d885c4e : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/01100a7702234a57b0871aaeef6b9748 : SUCCESS in 45m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/386f4352fb0444fd996e76e02f0a524b : SUCCESS in 59m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f2128c72bcd4aa098cf853449023768 : SUCCESS in 7m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2a793144cc2a4961884df0073091cb67 : SUCCESS in 15m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d40d0ad9b467487a832b98e62d4305a5 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fbb3d3e29314c68a0e304dedb27cb15 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ff35af75a634a51bbca54537002a87b : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/53085f6120224eb7813081c725b73eab : SUCCESS in 34m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ed79841a0c3424891fe6a09c219bc57 : FAILURE in 8m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea465a2008fc459e9fd4692a475bbb60 : SUCCESS in 1h 12m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a0ead2a03d8432d93108a24d567cb55 : FAILURE in 1h 35m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6288071cae6e417399bbe0f094721c8d : SUCCESS in 1h 39m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/522728619441450aa6e030ab653626e2 : SUCCESS in 58m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b1e60ad6aad4962afc121e434381385 : SUCCESS in 1h 26m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddf4251cbd1e4f61863d60224ef8c919 : SUCCESS in 1h 21m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca7b16aedb8c4a14b4ac12820221a922 : SUCCESS in 2h 10m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccc4c8a7ac234404b1e40a136387ecf4 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":6},{"id":"790f4503e036d89ab121b98cccf1ae329761d652","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 04:03:52.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_XtremIO_FC/d8a892d/DellEMC_XtremIO_FC/1272 : SUCCESS in 1h 57m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_XtremIO_iSCSI/747e865/DellEMC_XtremIO_iSCSI/1294 : SUCCESS in 1h 40m 06s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"f060d13fd8412db0f21b19e441a4f6c109de84e7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 04:13:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/6/20210409104507/NEC-ISCSI : SUCCESS in 2h 06m 52s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/6/20210409104507/NEC-FC : SUCCESS in 2h 28m 36s","accounts_in_message":[],"_revision_number":6},{"id":"2f7d0d8f6dedafcc84e15c22152ded8f735acfbb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-09 04:26:54.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_PowerStore_iSCSI/a9af7c3/DellEMC_PowerStore_iSCSI/761 : SUCCESS in 59m 28s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_PowerStore_FC/084840e/DellEMC_PowerStore_FC/756 : SUCCESS in 52m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"22fc44136927b0200b3bba0d1cf86b82417a099a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-09 04:38:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/6/upstream-check/cinder-cDOT-iSCSI/93efe84/ : SUCCESS in 33m 00s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/6/upstream-check/cinder-cDOT-NFS/2a78f79/ : SUCCESS in 34m 04s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/6/upstream-check/cinder-cDOT-FCP/0a3f0ed/ : SUCCESS in 56m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"98b725dcb52e6fc06abd53c64bd408483afcd3dd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-09 04:55:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/6/upstream-solidfire-check/cinder-solidfire-iSCSI/041f97f/ : FAILURE in 1h 27m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"6bca6533e9a5391301710e618984aa55c0f915eb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-09 04:58:35.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-5/ns5_nfs : SUCCESS in 4h 39m 28s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-5/ns5_iscsi : SUCCESS in 4h 33m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1863fc03bb3981d9177ed10f8669a0c1d4de1262","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-09 05:10:08.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_Unity_FC/9346f35/DellEMC_Unity_FC/1101 : SUCCESS in 1h 18m 09s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_Unity_iSCSI/d8c3617/DellEMC_Unity_iSCSI/1038 : SUCCESS in 1h 43m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"65f2d09de72cb59ccd086a25abe7aa4bebb7eb08","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-09 05:21:36.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-6/Seagate-iSCSI : SUCCESS in 44:36.97\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-6/Seagate-FC : SUCCESS in 39:22.93","accounts_in_message":[],"_revision_number":6},{"id":"d27e3bd6b857b410dc1a6fbaf5780eb11d20cead","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-09 05:31:03.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-6/HPMSA-iSCSI : FAILURE in 1:01:42\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-6/HPMSA-FC : SUCCESS in 34:08.72","accounts_in_message":[],"_revision_number":6},{"id":"18460daa6a2d247dc54f8c7977a77611683d73ea","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-09 05:37:46.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_SC_FC/e159a8d/DellEMC_SC_FC/603 : SUCCESS in 2h 10m 25s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_SC_iSCSI/e370f13/DellEMC_SC_iSCSI/600 : SUCCESS in 2h 02m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"235c907b286c10303f7562ce44880f213ab72bf5","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-09 05:51:24.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-6/PVME-iSCSI : SUCCESS in 58:25.17\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-6/PVME-FC : SUCCESS in 56:37.86","accounts_in_message":[],"_revision_number":6},{"id":"adc8ba480c22356dd973ee24f8c98e9d9b357501","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-09 05:53:43.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-6/ns5_nfs : SUCCESS in 4h 43m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-6/ns5_iscsi : SUCCESS in 4h 32m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"5bf8c27c86bad26d230fe14b762ebd148d32a682","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 06:09:45.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/6/check/DellEMC_PowerFlex/cc4bdc2/DellEMC_PowerFlex/749 : SUCCESS in 2h 42m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"cce049b8aad9a2b2692521ff2a00643928a6d926","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 06:22:09.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_VNX_iSCSI/ebe05c1/DellEMC_VNX_iSCSI/729 : SUCCESS in 2h 55m 01s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_VNX_FC/cdae718/DellEMC_VNX_FC/778 : SUCCESS in 1h 30m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"17cbd718536a0c71c4ad821e6485935862d1bd76","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-09 18:43:13.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/6/check/tempest-dsvm-full-focal-py3/5af5b28/ : TIMED_OUT in 3h 08m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/6/check/devstack-plugin-ceph-dsvm-focal-py3/bcc6c7f/ : TIMED_OUT in 2h 37m 50s","accounts_in_message":[],"_revision_number":6},{"id":"d5a027efc152ed0dec20deb8c7f1bb3d3ec546ff","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-09 22:26:41.000000000","message":"Patch Set 6: Code-Review+1\n\nI left some comments on the launchpad bug explaining the workflow of this, I\u0027m not sure if cinder people would like to update the commit msg or if it\u0027s OK as it is now. \n\nIt solves the encryption header issue.As far as I test it. Create a vol-luks1, attach it and create some data then a snap-1 from vol-luks1 and finally create a vol-luks2 from snap-1. I was able to read the data from vol-luks2.","accounts_in_message":[],"_revision_number":6},{"id":"fa9626d56691bf9e15cd8b9f72dd01854a80d461","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-10 19:40:48.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/23/784623/6/check/DellEMC_PowerMAX_FC/142230a/DellEMC_PowerMAX_FC/695 : SUCCESS in 5h 35m 32s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/23/784623/6/check/DellEMC_PowerMAX_iSCSI/831d626/DellEMC_PowerMAX_iSCSI/647 : SUCCESS in 5h 11m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"b29ade467801efad9f9492a7fd389b718153df5d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-19 17:20:21.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"99003b9ab4074245cabce03b183e18aaf960cf81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-19 20:09:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9444776ad4d10176289dc3d11bdc673183771710","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 20:34:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00ef85ac73904a5cb61e2f579b158047 : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bae5e5430c744542ab05af4601c247bf : SUCCESS in 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ab85dcdffec022bc6e5095549f91e36e20733046","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-19 21:07:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/7/iscsi/ : SUCCESS in 56m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/7/smb/ : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":7},{"id":"eb94c4b1f61c650d46e2857b86dc398b4ed6b757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 22:12:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fd45c59a884471fbe2bc69264a04fbe : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b946fc67a3364b079c1e77c412124330 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e005a2c13449e6aec61e94cf136870 : SUCCESS in 16m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63d99b2a0732403ba7ec7786787a2f4c : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eff6bd7518704e23a9fe59aa8ef130a0 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc4d46b621d4122acb7687ea85a0e84 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/b853f61b223c4ec589cde67875c6d3c2 : SUCCESS in 1h 14m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef6ec499d85c4fbfaaa305e912c489e1 : SUCCESS in 55m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba379f02c9f544558cc8ab204aedac7a : SUCCESS in 6m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6c833504f96460594142d2eb91ffa94 : SUCCESS in 19m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0921c3e6fb174311ad3b6c5f1d92e32f : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc349bfaea37410f9089a6e63b2611de : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/201ecd7c4c17404986fdecbe4e4cabfd : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82692781c8c04495958e66eea2ecd087 : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4fdbf006d47b4236a38829755c78a4dd : FAILURE in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1da64a82d26c457583a9b712eea7e063 : SUCCESS in 1h 05m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c3b470faeded44849106a704b0f8632f : FAILURE in 59m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8af38e0fd21044b8a6a0d91546e18aa4 : SUCCESS in 1h 06m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d48d4c4f0fb04cb094903cc3ed50d55e : SUCCESS in 1h 00m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67323cba44174500b5cbfdc726209449 : SUCCESS in 1h 10m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c7fb8baa724542c39f82f0f1d15afe98 : SUCCESS in 1h 04m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e19423468be24cbba39c3ae434d2cae7 : POST_FAILURE in 1h 59m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cac84bccf90f49de8feaef78cf1c42b3 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":7},{"id":"ee924dba6d6772419379f3304171d0e60e8c0c70","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-20 02:04:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/7/check/dsvm-tempest-cisco-zonemanager-job/1cc09fc/ : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":7},{"id":"b0acdc1034c037c674067f0514bc88aa0d4bfcd1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-20 02:27:19.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/689563d/ : FAILURE in 1h 45m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a746e2a/ : SUCCESS in 1h 53m 33s","accounts_in_message":[],"_revision_number":7},{"id":"69c0cc7f27a505cfbe288f9a44cf5c30af14f5cc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-20 07:44:34.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-7/ns5_nfs : SUCCESS in 5h 27m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-7/ns5_iscsi : SUCCESS in 5h 16m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"1d91a74b58918761054ac5a1220b2595489c5919","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 10:26:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"7017f9287a112bae601092ad6c5d17fde1ef7674","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 19:29:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/7/20210421020652/NEC-ISCSI : SUCCESS in 2h 21m 30s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/7/20210421020652/NEC-FC : SUCCESS in 2h 23m 01s","accounts_in_message":[],"_revision_number":7},{"id":"1aacac65d5a54e35b0042cf6795737a245fe10f4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 20:49:31.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/7/check/DellEMC_XtremIO_FC/23e853a/DellEMC_XtremIO_FC/18 : FAILURE in 2h 06m 30s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/7/check/DellEMC_XtremIO_iSCSI/41e2de1/DellEMC_XtremIO_iSCSI/18 : SUCCESS in 1h 54m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"9c8645bc90eb685b079f7be7c5128e725c247bc9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 07:49:30.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/7/check/tempest-dsvm-full-focal-py3/4e28c86/ : SUCCESS in 1h 30m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/7/check/devstack-plugin-ceph-dsvm-focal-py3/5a99dab/ : SUCCESS in 2h 00m 23s","accounts_in_message":[],"_revision_number":7},{"id":"c7edbe90d699670a5a93a434564474f63bf20530","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-22 15:29:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c6e39f5be32cd23a90f80efddeb0682834a2d48b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-22 16:01:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/7/check/cinder-storpool-tempest/e8157cc/ : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":7},{"id":"c52985a0cc8f4b26a6a9d44fb5a2b221264af1c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 16:04:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/444851e4d51b45ab83d7b801a88ec100 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f57d02a77cf40159eeebe88b66383f9 : FAILURE in 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"220b3a391e85fd04397ff5856188644284043293","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-22 16:44:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/7/check/dsvm-tempest-cisco-zonemanager-job/41f0692/ : FAILURE in 1h 09m 38s","accounts_in_message":[],"_revision_number":7},{"id":"4778e75b083a4d3523b8d7993ed8e2656ec6bfd9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-22 16:45:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/7/iscsi/ : SUCCESS in 1h 13m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/7/smb/ : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":7},{"id":"3d393de44266c02999b4a68fec5d74c45f246cfa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-22 16:59:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7ec7c22/ : SUCCESS in 1h 26m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/257471b/ : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":7},{"id":"95d753ae69a9de46ba7c28d6842e7e37f01a043f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-22 18:28:14.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/013c890b0e2545e185157ea266f4554f : POST_FAILURE in 1h 41m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/20f82f0793c544ba909ab20715fae293 : POST_FAILURE in 1h 26m 49s","accounts_in_message":[],"_revision_number":7},{"id":"6d5a199ac3872755d9fa6f7aa1a0972240cd2241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 18:56:35.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7acfe2724bdc47a5adccc3b99bd585e4 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b3a9469f5548ab9c304f03725ae5d1 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45ecfe7d646840b08433203091675404 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c81eb661274918b64cf983ff9ff69f : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/573df05a99c14b73816e03ecc86479f0 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3167d557504941f9ac8b6df8c69555c7 : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/f051ead280114ac585c5af4d30c28b8e : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48cadf9f940a4825bf95e15bf03e2902 : SUCCESS in 1h 03m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b8f194791dd4d90afe9de43d0784bcd : SUCCESS in 6m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6bcd3159aea64ff48a66e25e92207a85 : SUCCESS in 13m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/181a3937d146406db46c64153099ded4 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bea785087b64370956c7fa2d1ddc943 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c2723eb47f694c8aad04acac96784473 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/43caff5a94d4447caf131ac54b8c7d68 : SUCCESS in 41m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f14dae47880e4df8af8629f47440b582 : FAILURE in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71800bcb84cb4a43b72ec36bdaf7cb4e : SUCCESS in 1h 28m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/805cd059b7f741c48907665f2eb4d223 : FAILURE in 1h 27m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff09fad91ba24bd0af84e02c7da60532 : FAILURE in 1h 43m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f91488568a814543bd092f4c6f3e9e2f : SUCCESS in 1h 25m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f466f094c026453bb8f896bb62ef3edc : SUCCESS in 1h 29m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/352ddae8eb274dc3a4dd8601afabac49 : SUCCESS in 1h 25m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b18e455ff2044d52a9000ed6c89e653a : TIMED_OUT in 3h 03m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd31e4cf46454b64ab34139b403c78fa : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":7},{"id":"fd7bfa86bff0a98a2d7a2c2440176ea536ae5afa","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-04-27 02:35:40.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"5e30c61db015d89dc5c4616c35f36b7779e41789","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 02:58:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9baf1c3a17604dc2b599fa5af84394b1 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/071d1dd52c4f49e989889779953ed583 : SUCCESS in 21m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ff6ce091d9b395bfcb74e0584ddaf778a3cf1a3d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-27 03:06:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/7/check/cinder-storpool-tempest/9cc7473/ : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":7},{"id":"0f72d57ccfe2cf36d6a7cb5b3c252c7fceb34b86","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-27 03:51:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/7/check/dsvm-tempest-cisco-zonemanager-job/65b1a0f/ : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":7},{"id":"3ac9ac2bc82782d2528225724e3dffa250f313d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-27 04:08:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/39092a2/ : SUCCESS in 1h 29m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c8fb741/ : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":7},{"id":"d770104ea367eca154ba9f50fd5959d3ae1f22d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 04:23:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/364b56b496784f87a15705fb74d11d97 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b7a9ec172c4b49b905eb0ddf529359 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe02a25fd722481e88a223740ff6e2de : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1f283b8d6dd477e9fd7289da7385ba0 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/899f7fccf13c4a46979aa3d76f1fb97c : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0245f8c11143429ae5cce51e3f7cd0 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/5fab9dc0e40e44048f75813aa3b2a9a5 : SUCCESS in 58m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61a1995fec234ff89fb6bf96c89d8333 : SUCCESS in 1h 03m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfda32151a974ab495041e8a0511aeb7 : SUCCESS in 6m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1d8c4bb6fff43bc930bdb4d8598d341 : SUCCESS in 13m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59c3a30e850d4b06b83f9c49d8140340 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72bdfc6c5e3048a5abbe77be74f83448 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a14ed9151e44870a4329533e93d37a0 : SUCCESS in 8m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/983bf7a8540947c296a827f8d3627a40 : SUCCESS in 37m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff499c5dfd9a4704af2f49db3a5eae69 : FAILURE in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91396f3e2f0e40cda805333ff9481d66 : SUCCESS in 1h 11m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2116cf816d714e4bafad39d0d8686f0f : POST_FAILURE in 17m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1026579716834e88a4b5f6021ee92125 : SUCCESS in 1h 19m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6656dc3b8c842f683ed6544523885aa : SUCCESS in 1h 06m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71e80713a6dd41769559a6e711919f42 : SUCCESS in 1h 16m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a38189f1c33a4d34b95bb649f73c355a : SUCCESS in 1h 08m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf98ed6e23fe481586ffb4cc4aec83ee : SUCCESS in 1h 45m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f8243272d4c4fd4aba8e238ef0fa740 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":7},{"id":"7f5e215213bfe5493974bd147fa9022fb95c9e92","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-27 04:24:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/7/upstream-solidfire-check/cinder-solidfire-iSCSI/64b48a4/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"65a2b1916e0c8513c4231298a8000878f3b5bd84","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-27 04:37:01.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/7/check/tempest-dsvm-full-focal-py3/3a32561/ : SUCCESS in 1h 28m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/7/check/devstack-plugin-ceph-dsvm-focal-py3/932f33b/ : FAILURE in 48m 31s","accounts_in_message":[],"_revision_number":7},{"id":"507037d124a22586f254dd023021764986a00b70","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-27 04:47:27.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/7/check/DellEMC_PowerFlex/5c7dcf1/DellEMC_PowerFlex/19 : FAILURE in 23m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"4116d0344059213f4892969677b82e0981d10b26","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-27 05:10:47.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/7/check/DellEMC_PowerStore_iSCSI/2892b46/DellEMC_PowerStore_iSCSI/6 : FAILURE in 46m 53s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/7/check/DellEMC_PowerStore_FC/2a045cb/DellEMC_PowerStore_FC/7 : SUCCESS in 45m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"11fab2f4ed70509871062c48b55f4c5e464a7925","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-27 05:36:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/7/check/DellEMC_VNX_iSCSI/e268f46/DellEMC_VNX_iSCSI/34 : FAILURE in 1h 12m 34s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/7/check/DellEMC_VNX_FC/f563376/DellEMC_VNX_FC/37 : FAILURE in 55m 52s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"1d3acef1c6b02804d8907460697e195a80f0d33f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-27 06:05:06.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/7/check/DellEMC_Unity_FC/af16ba1/DellEMC_Unity_FC/53 : SUCCESS in 1h 18m 32s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/7/check/DellEMC_Unity_iSCSI/a06ea36/DellEMC_Unity_iSCSI/52 : SUCCESS in 1h 41m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"ce753fdecea1ae018bf25706423648f7a8907d80","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-27 06:22:22.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-7/Seagate-iSCSI : SUCCESS in 45:49.87\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-7/Seagate-FC : SUCCESS in 40:34.79","accounts_in_message":[],"_revision_number":7},{"id":"adb84ac29527d9cf9f1bd00dfc9124ff93425037","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-27 06:31:35.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-7/Lenovo-iSCSI : SUCCESS in 50:23.61\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-7/Lenovo-FC : SUCCESS in 46:31.61","accounts_in_message":[],"_revision_number":7},{"id":"28fc209560966b68667f2c0e94774cbb697f3b68","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-27 06:31:58.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-7/HPMSA-iSCSI : FAILURE in 1:04:03\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-7/HPMSA-FC : SUCCESS in 33:24.94","accounts_in_message":[],"_revision_number":7},{"id":"aa896b2163fb9a3d1fa8b2f957bdbec5539c8404","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-27 06:32:32.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/7/check/DellEMC_XtremIO_FC/3e0016a/DellEMC_XtremIO_FC/62 : SUCCESS in 2h 08m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/7/check/DellEMC_XtremIO_iSCSI/54c7d17/DellEMC_XtremIO_iSCSI/62 : SUCCESS in 1h 54m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"35540ac3bdfadb5df3aaf4b2429e61c75bf7fc4d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-27 06:55:21.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/7/check/DellEMC_SC_FC/5a849c1/DellEMC_SC_FC/38 : FAILURE in 2h 13m 07s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/7/check/DellEMC_SC_iSCSI/0a415d5/DellEMC_SC_iSCSI/35 : FAILURE in 2h 13m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"a6eae6089354c3e8602b004d03d440547b9d2238","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-27 06:55:28.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-7/PVME-iSCSI : SUCCESS in 1:03:09\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-7/PVME-FC : SUCCESS in 49:06.56","accounts_in_message":[],"_revision_number":7},{"id":"c13538eb1156a7a36d9a690cdcc8953b73b6586a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-28 05:33:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/7/upstream-check/cinder-cDOT-iSCSI/c97e681/ : FAILURE in 15m 26s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/7/upstream-check/cinder-cDOT-NFS/287119d/ : FAILURE in 16m 09s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/7/upstream-check/cinder-cDOT-FCP/449b68a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d62ac35792c7d782b35b2a2055a810a369ad0228","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-04-29 05:03:02.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/784623/784623/7/check-not-vote/ext-cinder-zuul/62ba912 : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":7},{"id":"1031f2c5cccb097761aaf407b2370fcbd97448d6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-03 13:17:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nHi Sofia, thanks for the fix.\nI\u0027ve a few comments inline although I\u0027m not 100% sure i understand the fix.","accounts_in_message":[],"_revision_number":7},{"id":"1218bf67ba1c3fdd6fed275ddea7ba2c5a1ca9e7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-03 13:51:08.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c15ec92917d33ba59015adec757a39dc292355a1","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-05-03 15:03:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"97546b0b203f0be517e06ba6005db57cf2090f02","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-05-03 15:51:00.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nHi Rajat, let me know your thought","accounts_in_message":[],"_revision_number":7},{"id":"2f8472f92b0c7d278a84adafdf7e577aaa562aa6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-03 17:06:17.000000000","message":"Patch Set 7: -Code-Review","accounts_in_message":[],"_revision_number":7},{"id":"f4f11048f5b0d7218e520cd1fe6353029dd71084","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-05-05 06:37:47.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nHi Sofia, I Checked the logic and seems like this one is written when the destination volume is bigger than source volume. I\u0027ve added one more launchpad link inline in which this problem persists even when we create the same size volume as the source volume.","accounts_in_message":[],"_revision_number":7},{"id":"54092041f08a7c701a8225ecdd9a41553591e23a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-05-05 23:57:54.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"781047e9273c5d6afe5c1b32d16aa46fe7495873","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 00:20:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a496d466af224edfbe27d6a1416c442f : SUCCESS in 17m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f30fa182c9c40cda79f62f1990f40fb : SUCCESS in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"90287eb4080e5425a7d6273880476c7e41fbac4e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-06 00:31:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/8/check/cinder-storpool-tempest/73e41db/ : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":8},{"id":"4c3459e667b33b027c560e8629b993921a3cdac6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-06 00:47:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/8/iscsi/ : SUCCESS in 47m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/8/smb/ : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":8},{"id":"6bbeca6c86658c4a903b4ca226511b5da67f914e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-06 01:10:17.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/8/check/iscsi-driver-primera/c31cc19/c31cc19.html : FAILURE in 1h 10m 43s","accounts_in_message":[],"_revision_number":8},{"id":"ebbcc2f6dd6b196fbe5ec0c781f23707942bfd5c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-06 01:16:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/8/check/dsvm-tempest-cisco-zonemanager-job/ac15c65/ : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":8},{"id":"77ccee2ff8793a1d18ba7f3935c847a75b155160","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-06 01:31:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/8/check/dsvm-tempest-zadara-driver/9b8e303 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f2efae0a801a9f0183f70a72bfb4caba20ed2cfa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-06 01:35:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/556b446/ : SUCCESS in 1h 35m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/becb970/ : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":8},{"id":"6acad3f402b868b53f4a0a39b8edcc0d3896ee92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-06 01:36:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_Unity_FC/f55ac8d/DellEMC_Unity_FC/121 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_Unity_iSCSI/dd185ca/DellEMC_Unity_iSCSI/112 : SUCCESS in 1h 37m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"88968a93b60dfaddae522e6198fc785c6bc1dc49","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-06 01:57:58.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/8/check/iscsi-driver-nimble/a4229ca/a4229ca.html : SUCCESS in 1h 58m 31s","accounts_in_message":[],"_revision_number":8},{"id":"a1b8c11b54d16eacb460405f2b9ae511d6d0608f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-06 02:08:55.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_XtremIO_FC/566915f/DellEMC_XtremIO_FC/127 : FAILURE in 2h 10m 05s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_XtremIO_iSCSI/ba9ceb2/DellEMC_XtremIO_iSCSI/127 : SUCCESS in 1h 54m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"a0b61eb41af29b961e9de40fcf52b3a2da252594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 02:22:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cca8fa8d53e148dd84b11bd8fb6f3d34 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b8c1ddb539433fb82a1b80ede1fe8d : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e729afe6c2d1453fa086cc7979009b38 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb5aaa74e1ae4c23bb0a7b991faa5a2f : SUCCESS in 11m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cc8e1eccc1948ac9e6d001137e43619 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60140003656e43a8878b6a73d12c8367 : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/001ec972f51e4cd799fb78d5ebb4868f : SUCCESS in 44m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d0e876619344f1ba6142884cf6cd50a : SUCCESS in 57m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d8276685ad24da699feb92876380467 : SUCCESS in 6m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18859551e3224d58a85d2a139c48c632 : SUCCESS in 10m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77e2d9a8c4584b78b81dbbd6a8c17609 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a39f53369d5b4e9c9cbab7b80001a0d7 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a615222e0b74105b4a2c6a947e59bce : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0ffca6caa1614d69952ddf9293d52642 : SUCCESS in 43m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/109a2f8b9ccd4dcd8526e85cb0d5f630 : FAILURE in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/879be4198a7042dba2b08c8e11ae40e8 : SUCCESS in 1h 33m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6690a068f05a4017b32a4143c4755498 : FAILURE in 1h 34m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/479c0c9020b14fdea4cfb3a8bbcf672d : SUCCESS in 1h 45m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7eeeaabb10ea4d2190e4e636dbd1686e : SUCCESS in 1h 26m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ace72a320a74adb8df4e09526ad202d : SUCCESS in 56m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e732305b2de4883bfa6e2ec0cdcb670 : SUCCESS in 55m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e49d6520aeb64afcb69ffe9c43f7861d : SUCCESS in 2h 17m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a2a54d5d3ce485f8aba1fc2fc369e1a : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":8},{"id":"26952f89d63337d91eec18e981d32d815770c964","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-06 02:27:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bdbcce9/bdbcce9.html : SUCCESS in 2h 27m 53s","accounts_in_message":[],"_revision_number":8},{"id":"dee6d3acbce2c11a11ed9549177cf1fa7db973f4","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-06 02:33:15.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/8/20210506085808/NEC-ISCSI : SUCCESS in 2h 02m 36s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/8/20210506085808/NEC-FC : SUCCESS in 2h 35m 04s","accounts_in_message":[],"_revision_number":8},{"id":"aea9a0e4fef9043c228248d43357ed24de6f3d02","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-06 03:06:11.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/8/check/tempest-dsvm-full-focal-py3/f71382f/ : SUCCESS in 1h 34m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/8/check/devstack-plugin-ceph-dsvm-focal-py3/13985ae/ : SUCCESS in 3h 02m 27s","accounts_in_message":[],"_revision_number":8},{"id":"74eeb28d4197c0f4f39909f6694f40772cc13606","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-06 03:09:05.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_PowerStore_iSCSI/b6e1801/DellEMC_PowerStore_iSCSI/50 : SUCCESS in 45m 09s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_PowerStore_FC/2335149/DellEMC_PowerStore_FC/51 : SUCCESS in 46m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"ebc394e6378d4e0680fdd4198954c13175461155","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-06 03:19:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/8/upstream-check/cinder-cDOT-iSCSI/36bb547/ : SUCCESS in 35m 51s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/8/upstream-check/cinder-cDOT-NFS/d9c78db/ : SUCCESS in 36m 45s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/8/upstream-check/cinder-cDOT-FCP/285dd21/ : SUCCESS in 55m 56s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"c9ce9478d9facb307a6ab6d403ea52e293b7f799","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-06 03:19:28.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/8/upstream-solidfire-check/cinder-solidfire-iSCSI/02aa0f3/ : FAILURE in 55m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"2bad56c5d48316a361a4a03584d6bca45fbf6989","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-06 04:15:17.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-8/Seagate-iSCSI : FAILURE in 45:50.31\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-8/Seagate-FC : SUCCESS in 37:28.26","accounts_in_message":[],"_revision_number":8},{"id":"a0eb96ad2e6e6ff44292c7a83eb1a674d01a3163","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-06 04:17:03.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-8/ns5_nfs : SUCCESS in 4h 18m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-8/ns5_iscsi : SUCCESS in 4h 17m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"d45df6ccdc992d310817794e14c9d95d1c711278","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-06 04:21:47.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-8/Lenovo-iSCSI : SUCCESS in 47:13.94\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-8/Lenovo-FC : SUCCESS in 43:27.89","accounts_in_message":[],"_revision_number":8},{"id":"a1ea4347f3c744e267e0720ce0d7e503f16a6189","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-06 04:25:36.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_SC_FC/2dda73a/DellEMC_SC_FC/72 : SUCCESS in 1h 57m 48s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_SC_iSCSI/ac49dae/DellEMC_SC_iSCSI/69 : SUCCESS in 2h 02m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"530479afd0d6ef58b793d4711921356c8a08dfb7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-06 04:35:15.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_XtremIO_FC/8a9e5da/DellEMC_XtremIO_FC/128 : SUCCESS in 2h 12m 32s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_XtremIO_iSCSI/a47f134/DellEMC_XtremIO_iSCSI/128 : SUCCESS in 2h 02m 26s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"9ab798f2dc6eab1edfefaac9476acc1f1fb6f36d","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-05-06 04:37:56.000000000","message":"Patch Set 8:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-8/PVME-iSCSI : SUCCESS in 53:43.29\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-8/PVME-FC : SUCCESS in 53:27.02","accounts_in_message":[],"_revision_number":8},{"id":"760d3f4470bc267f54cc59713ee39626a6852d86","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-06 04:50:47.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-8/HPMSA-iSCSI : FAILURE in 1:26:29\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-8/HPMSA-FC : SUCCESS in 33:28.89","accounts_in_message":[],"_revision_number":8},{"id":"e9a66a23a58fd6a81eb35b61481fa41fa618280b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-06 04:55:04.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_Unity_FC/364407d/DellEMC_Unity_FC/122 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_Unity_iSCSI/7a2473b/DellEMC_Unity_iSCSI/113 : SUCCESS in 1h 31m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"8aa4671ceca3cdaf9766dc68d63183b6de0b8b30","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-06 05:26:51.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/8/check/DellEMC_VNX_iSCSI/ab347f6/DellEMC_VNX_iSCSI/76 : SUCCESS in 3h 04m 02s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/8/check/DellEMC_VNX_FC/0567278/DellEMC_VNX_FC/79 : SUCCESS in 1h 26m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"d0d10dcaad8d110bebfd3fd231507009aa67e6a7","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-08 06:49:19.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"61f735f35a48ff67436cc53f323ae201fdc85c36","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-10 06:10:37.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"16c4b35e3618bc983313e8260f23592e076bb445","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-17 17:25:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1b30b30d1fc8f77b66aca333f01436a4e5409050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-18 01:02:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0fcbaf3786a4a50f34926b3ce315b48c67b8db6f","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-18 01:03:36.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nhi, Eric Harney.\nthank you for your review.","accounts_in_message":[],"_revision_number":9},{"id":"3442e7920000ece5b8732d35fb44373cca3e26fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 01:24:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cd4c10b16a44371bd934282600a353a : FAILURE in 20m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/611ba908ae2542d281ac46b6574bfcfb : FAILURE in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"62440510d24ab8889ec12246a8f511c8f3b99be7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-18 01:30:03.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/9/check/iscsi-driver-nimble/982e8d2/982e8d2.html : FAILURE in 26m 30s","accounts_in_message":[],"_revision_number":9},{"id":"357037189b59bf4639c8fce4e7d0b936149200fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-18 01:33:39.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/9/check/cinder-storpool-tempest/01a0e5d/ : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":9},{"id":"daa1df4cbc886405502c91069ad76004b2cc93a5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-18 01:52:24.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/9/iscsi/ : SUCCESS in 48m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/9/smb/ : FAILURE in 49m 30s","accounts_in_message":[],"_revision_number":9},{"id":"8473422fb0d9d74d04e7c70d2d6acc99a233e95c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-18 02:17:00.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/9/check/dsvm-tempest-cisco-zonemanager-job/95982de/ : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":9},{"id":"f0294b21aeb3a11314fe1a4bd536070fa2296b40","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-18 02:18:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/9/check/dsvm-tempest-zadara-driver/0c2aa51 : SUCCESS in 1h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fd85bcc1ec5e876accd99de2bcd7379b86ef3fee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-18 02:42:06.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/9/check/tempest-dsvm-full-focal-py3/408b63d/ : SUCCESS in 1h 26m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/9/check/devstack-plugin-ceph-dsvm-focal-py3/1a73470/ : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":9},{"id":"80cbf9bf162afb0c8404948f5b6f784861fa877a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 03:00:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecc77e5abaf540d69bf46c9691509665 : FAILURE in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510f09ab689c4bbdb53ef16da9af2c68 : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eec1014f9fc4b63956698735d427395 : FAILURE in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca70e05e78243bc886f7cbadd0ee240 : FAILURE in 15m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de40eca06ecf46efbaa7ad97a6067714 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c649eb924a4e16bf4481f664f3be27 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ef0bf6279f57486ea0927e29daeba559 : SUCCESS in 1h 01m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6956908941184738bc0c75ab21aae1e1 : SUCCESS in 1h 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8bae2a479cf459f9bcb8ab73cc76e9c : SUCCESS in 5m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b91be0d2cd884a3193e7759fc438532c : FAILURE in 10m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e3a9d78742746efa94cf21e93329dcc : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04935f0d7cca4fb6afdd36e1cd72526d : SUCCESS in 6m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/155ff69c7ece428fa985d7a0408310af : SUCCESS in 7m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a5ba6d03cbd4086a42464ada82643a4 : SUCCESS in 30m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1739044890e148dfb9ed1d001fcd7379 : FAILURE in 9m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/90757d5237764fcd98bc1e2b94ca1078 : SUCCESS in 1h 09m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a38368ba19604e1492ec8fd85c3670d2 : SUCCESS in 1h 21m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b8d5176ff27b4de68f7dea839822f62a : FAILURE in 1h 11m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/362ab1c6aa1b4724ad29192ae294db28 : SUCCESS in 1h 13m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21aa9025da624a27954f3844a4644653 : SUCCESS in 1h 01m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc2ff3bc56f649dfa668817125f77ac2 : SUCCESS in 55m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4cd2ca149ae4696a4c92297b526b193 : SUCCESS in 1h 56m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64261438786944ba9649de8e40c44583 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":9},{"id":"2ae934b5710d207516fbf899983e66761307fb42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-18 03:21:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d7d1ac1c63bb582f13499c61e4e9d38628d793a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 03:46:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e88bd4bf67c74527b6a8b7662afaca67 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd296424b16742bc85b8697db1b19d56 : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d76841bfdf9ea1cbb17427afaf68706767079166","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-18 03:48:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/10/check/iscsi-driver-nimble/ebe22ad/ebe22ad.html : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":10},{"id":"4abb39b4747cf45aaad2f2b342de04bec6a91ed2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-18 03:52:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/10/check/cinder-storpool-tempest/ddb490e/ : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":10},{"id":"2f23adf0ef3d59098f33178a4f39b34f55d17c35","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-18 04:11:04.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/10/iscsi/ : SUCCESS in 49m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/10/smb/ : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":10},{"id":"d4444f2c5ffc5d0063cd3303377f3b410627c4b9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-18 04:32:44.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/10/check/dsvm-tempest-zadara-driver/25a4d5c : SUCCESS in 1h 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cd9e67b76d49aa22ed72a4a25944be65a8fed6b1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-18 04:38:49.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/10/check/dsvm-tempest-cisco-zonemanager-job/022ed36/ : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":10},{"id":"743eb8fe2ac13e2024804bf66336234eaa6fd5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 05:25:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7c52930891e408391d6e196532920af : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/493c51883cd04486b0c62302b05c2acc : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a7e3a8e83ad46e7adbe1148e0ad4eb2 : SUCCESS in 14m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c432394d0f774d73be72e26b13c62a66 : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9a0112033a14dfcab861aa085b1746b : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504c4c90a802458da36d4e3c2265f092 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ab754fe4337c4720aad5705d02f41d14 : SUCCESS in 57m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4eafabcb6c3a48569cda4637e2d9afb7 : POST_FAILURE in 1h 46m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7f82ae1696446abb8c2b5ccf5541439 : SUCCESS in 6m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9dfcc8acefc4be284243fec8276912b : SUCCESS in 14m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d53e62adbbb448f198ac0b78ffa43984 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/adf15b2a6ebb417d8742d27a53a366d0 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac2bf0caf57d4e1ebcafb8acec702739 : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/213c616023e74eec8dad5c099bd790f9 : SUCCESS in 32m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f364b4240ae47559f8af952573fc45e : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6d4b56d373f74c14ac19bd418a542c8c : SUCCESS in 51m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e22772ba4d2c49218125c0d75f85ca66 : SUCCESS in 1h 23m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ad7332610c2f4dbaaf8d73ccd8b82225 : SUCCESS in 1h 29m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f0e1a1cfbfa94efaabbcde792799fc81 : SUCCESS in 1h 07m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aecd7fd9b0954fc8bcf8932d74410aad : SUCCESS in 1h 06m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89b364a624244fb19454e5d547a3b76d : SUCCESS in 1h 19m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e0ea64909dcb4278a69de565723e4611 : SUCCESS in 2h 03m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cf5098c7f49435182d7b51be71e81b1 : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":10},{"id":"e6ee878bf72f4ec1c74d572abd5ddf29606dde37","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-18 05:41:04.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8755dae56bd40c3959b80b9a5109d80af1142252","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-18 05:45:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/96140e8/96140e8.html : SUCCESS in 2h 23m 28s","accounts_in_message":[],"_revision_number":10},{"id":"0b455ede3ebeddc8e01fd6ebc476e3fc8a0e89e4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-18 05:55:07.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/10/check/tempest-dsvm-full-focal-py3/7212afd/ : SUCCESS in 1h 26m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/10/check/devstack-plugin-ceph-dsvm-focal-py3/d674c86/ : SUCCESS in 2h 29m 14s","accounts_in_message":[],"_revision_number":10},{"id":"5549e5d0efca487d154ed2c4629f805603769dc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 06:06:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac08fb07d5dd48d797ad567516f44cf3 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e32880be6e24b98804514699ba6f39a : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2d6675e9711f3ca069e46022ee2583d4def92fd5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-18 06:26:34.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/75657aa/ : SUCCESS in 2h 51m 40s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3f77a9b/ : TIMED_OUT in 3h 04m 26s","accounts_in_message":[],"_revision_number":10},{"id":"0d344d8a59ecbee0c2f7f530948f74bce1acaf86","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-18 06:33:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/10/iscsi/ : SUCCESS in 48m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/10/smb/ : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":10},{"id":"d943e5427b8f11e79cb521d12ad62500a92708cc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-18 07:00:59.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/10/check/dsvm-tempest-zadara-driver/b13451c : SUCCESS in 1h 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"96f9984e0e826d4e1008412aa30b7b52950966a8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-18 07:01:11.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/10/check/dsvm-tempest-cisco-zonemanager-job/5b1f7a6/ : FAILURE in 1h 14m 48s","accounts_in_message":[],"_revision_number":10},{"id":"ad58101b9ff8a7d8663cbb332bc39fa12ab923ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 07:25:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f59862520114da48ef47f0606821b20 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac439b32dbd6426bb60aeb0ef4352c2b : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f67e43c96a4e34a6ca8ae7fe939748 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38c67483c0b745f49f5a206faf927769 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d668098801f9481183919a4a8156920e : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e24298dd19e435d9c8d7b01efafdde4 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/65e8d7e155bc43288a3b9c6816c0a0f4 : SUCCESS in 1h 03m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45bc3ff442d840bd90b0a6ae25a6ba0f : SUCCESS in 1h 04m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff9eab7fe5f24a0b97570a676757a579 : SUCCESS in 6m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0b9c2b70738c4aa1803e083763e0aaba : SUCCESS in 11m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eefc12169e1e4b8d9102c96dfaad827a : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ed54f3b1f304b7795bdf255ef21fdf1 : SUCCESS in 5m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/935e54f5b38f4ba1ae1fdedc78c2f7c4 : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd58dd8aece94944bc15e10ab41dc728 : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/79d0a63aa3c040278f0ba29e484e046d : FAILURE in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1911baa7177f4d8b80df550ea454c1cd : SUCCESS in 1h 12m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/83fe6959b02d43da864d87ca0482f1dc : SUCCESS in 1h 16m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aea2077499004190b8ec0830b9bb78c0 : SUCCESS in 1h 28m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/261358b0e89d4763bb3a438b014562ba : SUCCESS in 58m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/198349fdee4e43a49f027371b9b774e8 : SUCCESS in 1h 06m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/09f25403f03b45429442ec3880d09558 : SUCCESS in 1h 17m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c7737625719497b841b0046fe9400c0 : SUCCESS in 1h 39m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/132e5c06d2ee4e528975acd78ea8e208 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":10},{"id":"6f869a1ed40ef0d30f5330de28360dbcb02026c6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-18 07:25:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/10/check/cinder-storpool-tempest/d3d0e29/ : POST_FAILURE in 1h 42m 52s","accounts_in_message":[],"_revision_number":10},{"id":"49899bf06d40284a2e18d561d0fcbde92c1f69e4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-18 08:13:32.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/10/check/DellEMC_PowerStore_iSCSI/b9406aa/DellEMC_PowerStore_iSCSI/96 : FAILURE in 45m 43s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/10/check/DellEMC_PowerStore_FC/7f9ac4f/DellEMC_PowerStore_FC/97 : SUCCESS in 47m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"73cddbcc898699ac1a7b5948082d6e7ba7f9617c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-18 08:32:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/10/upstream-solidfire-check/cinder-solidfire-iSCSI/e1e075b/ : FAILURE in 55m 11s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"b1228534e6978e754de84a17ccae57af313175ba","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-18 08:34:15.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/10/upstream-check/cinder-cDOT-iSCSI/d7d7e42/ : FAILURE in 38m 20s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/10/upstream-check/cinder-cDOT-NFS/4f26316/ : FAILURE in 11m 16s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/10/upstream-check/cinder-cDOT-FCP/62effdb/ : POST_FAILURE in 1h 06m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"5f499fc005e254b1b903c84fc373c2ea17a16eeb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-18 08:47:38.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/10/check/DellEMC_PowerFlex/9723f37/DellEMC_PowerFlex/105 : SUCCESS in 1h 21m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"61a1b70335ecf30c9d97a46bc9c9c7cd83c42c81","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-18 09:19:00.000000000","message":"Patch Set 10:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-10/Seagate-iSCSI : SUCCESS in 45:34.34\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-10/Seagate-FC : SUCCESS in 38:22.23","accounts_in_message":[],"_revision_number":10},{"id":"f7c5d9590ceb63519b462efe181df321176743cb","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-18 09:31:21.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-10/Lenovo-iSCSI : SUCCESS in 50:08.15\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-10/Lenovo-FC : SUCCESS in 45:57.97","accounts_in_message":[],"_revision_number":10},{"id":"7aec3ad9b4b27c2736d4b2c912cbc9330d634ab4","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-18 09:44:49.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/10/check/DellEMC_SC_FC/a98b2b7/DellEMC_SC_FC/112 : SUCCESS in 2h 18m 33s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/10/check/DellEMC_SC_iSCSI/bdd71f9/DellEMC_SC_iSCSI/109 : FAILURE in 14m 41s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"c058443fd818c3966a3526a61e33238b167affd5","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-18 09:54:59.000000000","message":"Patch Set 10:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-10/HPMSA-iSCSI : FAILURE in 1:26:44\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-10/HPMSA-FC : SUCCESS in 33:33.37","accounts_in_message":[],"_revision_number":10},{"id":"98db635aad248091e010502ea444042d5cfe4ef5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-18 10:34:53.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/10/check/DellEMC_VNX_iSCSI/d081b90/DellEMC_VNX_iSCSI/117 : SUCCESS in 3h 08m 36s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/10/check/DellEMC_VNX_FC/cb0f2de/DellEMC_VNX_FC/123 : SUCCESS in 1h 29m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"0212eb2b95c49bcaab97d137d20f256a880ea4ec","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-18 10:54:03.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-9/ns5_nfs : FAILURE in 9h 51m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-9/ns5_iscsi : FAILURE in 9h 51m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"465691015ee788380462a6552f5792f44f29ec4d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-18 11:25:50.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-10/ns5_nfs : FAILURE in 27m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-10/ns5_iscsi : FAILURE in 29m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"38191370163cce06b39752da079b793c79c73795","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-05-18 15:08:29.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2de92bb0908f6cf3ced40c8713601befcfb7c0ad","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-18 15:52:51.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-10/ns5_nfs : FAILURE in 26m 42s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-10/ns5_iscsi : FAILURE in 26m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"2b1675ec9f9095ca191fdd31a12432c0b372397f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-19 01:44:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f331c37c9029f793522912d51d13ab7ed02f475e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-19 01:45:23.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nhi, Gorka Eguileor\nthank you for your review.","accounts_in_message":[],"_revision_number":11},{"id":"42fd5cfb2fe035044c384f56aa2c8f1c9045a4fe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-19 01:53:57.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/11/check/cinder-storpool-tempest/73e56cf/ : FAILURE in 7m 33s","accounts_in_message":[],"_revision_number":11},{"id":"0c052b71bc562cd1bf81f593ff28e1c2bc23fb84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 02:06:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/517562d446fb46d1bb2dcd763894dec4 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16661978f31a413597c46883a374a95d : SUCCESS in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"66740c1305027fe3c517a737b9eaf4f867453058","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-19 02:08:52.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/23/784623/11/check/dsvm-tempest-zadara-driver/1331774 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"305c919ed4707169a4e55429cb98638236e62d5e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 02:11:05.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/11/check/iscsi-driver-nimble/29de06e/29de06e.html : FAILURE in 25m 17s","accounts_in_message":[],"_revision_number":11},{"id":"8f7acac26a6e762535e56b4e5b831afca3985f8d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-19 02:29:26.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-11/ns5_nfs : FAILURE in 26m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-11/ns5_iscsi : FAILURE in 26m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"c2884ba18cbf2ab454759d82b3d2b079460fcb13","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-19 02:45:10.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/11/iscsi/ : SUCCESS in 59m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/11/smb/ : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":11},{"id":"def1f68fc5a4b05ae972fcdff98f429d41807eaa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 02:56:52.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/11/check/iscsi-driver-primera/e914392/e914392.html : FAILURE in 1h 10m 47s","accounts_in_message":[],"_revision_number":11},{"id":"082a78c36589476595e436f228ab644e8eb48377","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-19 03:03:19.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/11/check/dsvm-tempest-cisco-zonemanager-job/74243f9/ : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":11},{"id":"4231998e42c7088bac175316fcb67a9107f7f709","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-19 03:23:55.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/eb8e8c1/ : SUCCESS in 1h 34m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/11/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f2da64e/ : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":11},{"id":"f15a91914a907ff1f4c6306c8ac7a24686edc865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 03:59:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc125c6282384dcb886bdf8940c4caff : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1afee9136a47c2818b39a76af934e4 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee56f53d9409431ab5d275c7a72ac438 : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b25753a4ef04fc38817d4e037378912 : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d02039accb4e4197b33dc0933576db05 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d7c685b75f444785acda9bd33a3927 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3ed9a3b390024e82a63da1b9223a7a92 : SUCCESS in 1h 01m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa980822f7da4ab9a654e47689765fb9 : SUCCESS in 1h 15m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fdaeed5a3674538a662ea6a76ac8134 : SUCCESS in 6m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/547d47555c434761ba6b55d381c5e231 : SUCCESS in 13m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77ff6e35dd8948d4982ed6f5ee712fd5 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76fc2cf9f5ad403abab933914a54ec25 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7e5d2825c99448aca75b66586abc2aba : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5da67f2f150d4c78be7234b252931d58 : SUCCESS in 38m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc09653ef2364646a15917b895c21ddf : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/00a740f75f5c472f8ce44afa9d2afa4b : SUCCESS in 1h 13m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e34e8696e84499cb093924d35d46cbb : SUCCESS in 1h 24m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f373a82fd07d478ca1bf96271631d255 : SUCCESS in 1h 26m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc5b69795210430ea16c689e4f36d6ce : SUCCESS in 1h 18m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4948765f935f4224bb1a0ff746095faf : SUCCESS in 1h 16m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fd7058e0b8a84412adac953e783a7dcc : SUCCESS in 1h 16m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/602feb3c52164605872f2fcf5c93c782 : SUCCESS in 2h 09m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f850e29f1674ffcb77a8b0f20361427 : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":11},{"id":"f032fc257055b529feb5f3642f385ef3b7182a39","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-19 04:02:06.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/11/check/tempest-dsvm-full-focal-py3/59cb4b8/ : SUCCESS in 1h 17m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/11/check/devstack-plugin-ceph-dsvm-focal-py3/baabbe5/ : RETRY_LIMIT in 17m 38s","accounts_in_message":[],"_revision_number":11},{"id":"597ab5037bddb95eba5ce665d3dd242db4b293f0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-19 04:12:53.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/11/upstream-solidfire-check/cinder-solidfire-iSCSI/e8cdc4d/ : FAILURE in 11m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"e3975d5a37df7a917cd87fd86b8b208214e8b60e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-19 04:13:19.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/11/upstream-check/cinder-cDOT-iSCSI/57dcdc4/ : FAILURE in 12m 19s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/11/upstream-check/cinder-cDOT-NFS/8c1a4fa/ : FAILURE in 10m 58s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/11/upstream-check/cinder-cDOT-FCP/8e2c0bd/ : FAILURE in 12m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"88b7a6b6dd42927725fdead81f3ddfea7fc69a65","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 04:26:36.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/941550a/941550a.html : SUCCESS in 2h 38m 20s","accounts_in_message":[],"_revision_number":11},{"id":"2fd3f6742ce4f3c0867d3f3611860c54fe5d172f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-19 04:50:18.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/11/check/DellEMC_PowerStore_iSCSI/f967cd0/DellEMC_PowerStore_iSCSI/99 : SUCCESS in 49m 24s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/11/check/DellEMC_PowerStore_FC/778d99e/DellEMC_PowerStore_FC/100 : SUCCESS in 50m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"7909d851b181e8d3cc28886d782f727450d521c7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-19 05:17:58.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/11/check/DellEMC_PowerFlex/d2a403f/DellEMC_PowerFlex/108 : SUCCESS in 1h 17m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"d4b7ca07d0c69e1a2571bf507a8a373c2e00b556","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-19 05:56:02.000000000","message":"Patch Set 11:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-11/Seagate-iSCSI : SUCCESS in 45:56.47\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-11/Seagate-FC : SUCCESS in 40:11.12","accounts_in_message":[],"_revision_number":11},{"id":"7bf6f4f194f2357e6f55a06b5daf57b7fc818b22","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-19 06:06:46.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-11/Lenovo-iSCSI : SUCCESS in 50:37.13\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-11/Lenovo-FC : SUCCESS in 46:39.30","accounts_in_message":[],"_revision_number":11},{"id":"16b7e5d18396081a58523b0bef18118966cd3213","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-19 06:31:03.000000000","message":"Patch Set 11:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-11/HPMSA-iSCSI : FAILURE in 1:27:04\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-11/HPMSA-FC : SUCCESS in 33:53.43","accounts_in_message":[],"_revision_number":11},{"id":"28cb0f4ef72868f3847bc290e74a17053c5627db","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-19 06:52:06.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/11/check/DellEMC_SC_FC/22cb2d9/DellEMC_SC_FC/115 : SUCCESS in 1h 58m 13s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/11/check/DellEMC_SC_iSCSI/e87b7a0/DellEMC_SC_iSCSI/112 : FAILURE in 2h 51m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"0d5a6687ec84f9848fa805275eea6b7b8f91fc2a","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-05-19 20:22:07.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9c356929aae9a514cd439c0ec88969da70e3fef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-05-19 20:26:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8a5bb11e34333484baccc322aadf17ae147e54b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-19 20:36:20.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/12/check/cinder-storpool-tempest/cdaba84/ : FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":12},{"id":"3f3ddb4bc28e3f4aaac81fdb3fe947f0a02fcf2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:48:38.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/591c08404d3143a0b2a79bc527b95eff : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03d35e0fba1042f3a322fa41d95332e9 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0667c97ac2ca3a493d8516ae984c390c3e388d39","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 20:54:34.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/12/check/iscsi-driver-nimble/e586328/e586328.html : FAILURE in 26m 34s","accounts_in_message":[],"_revision_number":12},{"id":"8d5287c2416d6825a40f0c8ea096dc38f1ba0894","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-19 21:27:44.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/12/iscsi/ : SUCCESS in 56m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/12/smb/ : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":12},{"id":"1fe505870af5e3fa956d963906ae5ecfa0bdadae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 21:39:10.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/12/check/iscsi-driver-primera/28c4e8c/28c4e8c.html : FAILURE in 1h 10m 40s","accounts_in_message":[],"_revision_number":12},{"id":"7792a93cdaa4fc749d3673f38834edfa80dd3997","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-19 21:47:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/12/check/dsvm-tempest-cisco-zonemanager-job/573a258/ : FAILURE in 1h 15m 33s","accounts_in_message":[],"_revision_number":12},{"id":"4844b0ffd2100c7690844386e01b8d1b245be60b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-19 22:18:06.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6454ccd/ : SUCCESS in 1h 50m 15s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/efb2ac0/ : SUCCESS in 1h 39m 15s","accounts_in_message":[],"_revision_number":12},{"id":"27b19d183bebafe31c2000f71ecdc71aa681e309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 22:26:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/068a5187e21f41b0ab8ac0c8a22cca91 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c19616d335d4a278165e7ea2a58de3f : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6a87b6e49024da89297baed8f601b5c : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff5d18588f8f4db8b1c13a0926b1b66f : SUCCESS in 9m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8a4787eed2a455f8a01bb50f3d27e43 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d108086d39425c9f76dd3c0632f22d : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/9168b1a2eecd45debe3bd6c37b323b82 : SUCCESS in 1h 15m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f84d0378c7864ec2bffef0e854ec2b80 : SUCCESS in 1h 15m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db9de6cdc3946fcb38dade4434f78f4 : SUCCESS in 7m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c097fe0005e743dda2664a4b72c2b30b : SUCCESS in 15m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72eeb1f958d5487aad31525f2bded145 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72aef7feff5a40429c5660613fb03c6d : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff551e15b183450d9a8d0a99dc823544 : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/530cc30059c94c68b013dbf177a4b822 : SUCCESS in 34m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04a3330a8e5844869338f8dc232c4289 : FAILURE in 7m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ecd0d7a409a497b9d82dbb1436317ef : SUCCESS in 1h 10m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/99c38443fd5b463980ca9ffa07aad37d : POST_FAILURE in 1h 57m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df2b2292eb974075b3880f696c8f9acc : SUCCESS in 1h 29m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6f8bd9b2d7f4852931911d6c240e9be : SUCCESS in 1h 09m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c9d5c22c0dd242618a2e27fa582a491c : SUCCESS in 1h 07m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/75ba576107e942da8a363004c36d051b : SUCCESS in 1h 07m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8154ec6c8f3d4212a6d36f6b1e21ff80 : SUCCESS in 1h 45m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2c1e7ed9e524652bf1c7e8b42367498 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":12},{"id":"01a1bea29e592f0ed1b4c52204fba954ca7075dd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-19 22:28:43.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/12/check/tempest-dsvm-full-focal-py3/e3de85d/ : SUCCESS in 1h 18m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/12/check/devstack-plugin-ceph-dsvm-focal-py3/17d6363/ : FAILURE in 1h 57m 20s","accounts_in_message":[],"_revision_number":12},{"id":"631f0261a4a03f01a522bdb2b6486d13cd99e817","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-19 22:38:50.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/12/upstream-solidfire-check/cinder-solidfire-iSCSI/194693f/ : FAILURE in 11m 11s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"d338befa3c8e58472ed87eedd4e6d95c63a9ff43","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-19 22:43:48.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/12/upstream-check/cinder-cDOT-iSCSI/9d4bf32/ : FAILURE in 16m 33s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/12/upstream-check/cinder-cDOT-NFS/3b39565/ : FAILURE in 11m 28s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/12/upstream-check/cinder-cDOT-FCP/f3bcb4c/ : FAILURE in 13m 14s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"960d831ef851da4ec3b9308aeb57a8eedf081214","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-19 23:30:19.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/7a071d7/7a071d7.html : TIMED_OUT in 3h 01m 13s","accounts_in_message":[],"_revision_number":12},{"id":"966960f4976df383647af0887bb25286bebc47c1","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-20 00:23:40.000000000","message":"Patch Set 12:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-12/Seagate-iSCSI : SUCCESS in 46:01.47\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-12/Seagate-FC : SUCCESS in 39:26.85","accounts_in_message":[],"_revision_number":12},{"id":"7c0e736091401f2f0441a41aecfd421c582ca929","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-20 00:37:43.000000000","message":"Patch Set 12:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-12/Lenovo-iSCSI : SUCCESS in 52:37.72\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-12/Lenovo-FC : SUCCESS in 46:35.44","accounts_in_message":[],"_revision_number":12},{"id":"0f16b34bd5c87fd56cda990f4573b62375b01d4c","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-20 00:59:21.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-12/HPMSA-iSCSI : FAILURE in 1:26:39\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-12/HPMSA-FC : SUCCESS in 34:32.33","accounts_in_message":[],"_revision_number":12},{"id":"613b6d040da34fa6965c62ad001e9f31f723d8c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-20 01:20:07.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-12/ns5_nfs : SUCCESS in 4h 43m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-12/ns5_iscsi : SUCCESS in 4h 52m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"98f7e64fdc01db6318482369efb46c6f261407d3","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-05-20 01:20:18.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"91ce035c1964a325370496003eb555f1d02d55a9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-20 01:34:15.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/12/check/DellEMC_PowerFlex/ab42a2f/DellEMC_PowerFlex/110 : FAILURE in 2h 50m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"e64d58ed1401e3ef73e699c009f2b9904eaf9188","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-20 01:53:09.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/12/check/DellEMC_SC_FC/01874f2/DellEMC_SC_FC/117 : ABORTED in 3h 00m 02s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/12/check/DellEMC_SC_iSCSI/f8ceef1/DellEMC_SC_iSCSI/114 : FAILURE in 49m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"3ff52a68d5ac1cd8fc149c0afd25fb16d97b4609","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-20 01:53:10.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/12/check/DellEMC_PowerStore_iSCSI/32db2bf/DellEMC_PowerStore_iSCSI/101 : ABORTED in 3h 00m 02s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/12/check/DellEMC_PowerStore_FC/d19b2d9/DellEMC_PowerStore_FC/102 : ABORTED in 3h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"ded8f84050117680b2fc2d666e43e3b1e683c476","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-20 02:02:05.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-12/ns5_nfs : SUCCESS in 4h 44m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-12/ns5_iscsi : SUCCESS in 4h 49m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"bbbfc2aa004d9cbb5ac6edbb8d56f0446e32c535","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-05-20 15:21:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a0141596f755e860fdc7fddf982115ac20a15837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-05-20 19:44:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f4c22f8786aadd975d61e849c4aeefdba327c52e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-20 19:53:59.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/23/784623/13/check/cinder-storpool-tempest/3c494d8/ : FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":13},{"id":"09d9a8818eda98206c7fd4cfed01de4d9f62453c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 20:13:37.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/784623/13/check/iscsi-driver-nimble/6b96cc5/6b96cc5.html : FAILURE in 26m 34s","accounts_in_message":[],"_revision_number":13},{"id":"df38760499a0830e33eccc68a29c65ed6e6eb00f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-20 20:13:39.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/784623/13/ : FAILURE in 28m 09s","accounts_in_message":[],"_revision_number":13},{"id":"cca7c68c2d90ab71e92ce14aabf43cfb27e6e0c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:14:49.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c488f734a3994505814a9fcfc2330896 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3701ba86301f43779c00ec021c6aef51 : SUCCESS in 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9949820f5b173cc343b14e8744dde0afbcbef101","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-20 20:51:17.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/13/iscsi/ : SUCCESS in 1h 04m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/13/smb/ : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":13},{"id":"9f45fc0ac0c39c95f070a34b7cc939ce6d7c7aa5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 20:56:35.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/13/check/iscsi-driver-primera/cbb3374/cbb3374.html : FAILURE in 1h 10m 45s","accounts_in_message":[],"_revision_number":13},{"id":"cda407d4dcb96961aef4a2467da16cc619a19f75","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-20 21:06:40.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/13/check/dsvm-tempest-cisco-zonemanager-job/8a0e501/ : FAILURE in 1h 18m 29s","accounts_in_message":[],"_revision_number":13},{"id":"6983908d7ea4dba9813a538c70be7a0e68b2e7fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-20 21:33:56.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6ae8537f049c428ea241fef3cd4dc649 : POST_FAILURE in 1h 39m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/5e31d90f22234f5bb8e7d35cdeb7f395 : POST_FAILURE in 1h 30m 12s","accounts_in_message":[],"_revision_number":13},{"id":"52a8ca5cea93b67b2ef60bff612ef512ed6e7544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 21:37:59.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/064f42ae6f1f4eeabf5bf8bf56f60543 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b17ebb759140c4832039ceb14242d2 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/993672d002b34901bad0d116ee097926 : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a50ac9445e74ee8ba4cd992f49756fd : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6255308e49ee4276830564b3c5517875 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c50b921dca414190f4aa4fec966c4e : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/d07f38422dc643f1aa826ed1f824837d : SUCCESS in 58m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5eb93a26c6084a19b3660c48b8157b73 : SUCCESS in 55m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286d24bcb4a740fab04dc0baffde85fa : SUCCESS in 7m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/15859896692c44938d5946e75254542e : SUCCESS in 12m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dfb6489d3344cd988937643ad5ce377 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b6e3db6cbeb43d782329027ea02ea8b : SUCCESS in 7m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/912ca2eb7c414204aefec350f5b8623c : SUCCESS in 9m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7dad4c56b10045d2ae73e8b3c5ebf572 : SUCCESS in 30m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c40c1d97ce654b21bf168c19c927ea85 : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cfff0e5a40e840b88913921d3d615479 : SUCCESS in 1h 07m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/329ae919dab048bc9580703eedd9f36a : SUCCESS in 1h 30m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a111bf9f03044e8ab7da929e40f4929 : SUCCESS in 1h 38m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ee3939d1af446269d63ad7f2d5e4c85 : SUCCESS in 1h 01m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46b0d3fbbfbe4e4f8ccf19d69deb27f4 : SUCCESS in 1h 06m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5e6ee7a2d2e456ab1869df34d8d0549 : SUCCESS in 1h 41m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa0d756800c543e79aa161cc57d48a39 : SUCCESS in 1h 52m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd928cf5e0a249ef8344be21d25907f0 : SUCCESS in 48m 47s","accounts_in_message":[],"_revision_number":13},{"id":"6ba658ceedfd1714876dc36240f2fc7509bb4ecf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-20 21:38:06.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/438ce7c/ : SUCCESS in 1h 50m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/429138d/ : SUCCESS in 1h 30m 07s","accounts_in_message":[],"_revision_number":13},{"id":"e7b4c71b2f51bf2b5051140311a5b78d7a1760c4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-20 21:50:42.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/13/upstream-solidfire-check/cinder-solidfire-iSCSI/17b64a3/ : FAILURE in 11m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":13},{"id":"cdd8e626cec75ec7a110d61c6dad26185041c9f3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-20 21:52:14.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/13/upstream-check/cinder-cDOT-iSCSI/871b315/ : FAILURE in 11m 48s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/13/upstream-check/cinder-cDOT-NFS/c873392/ : FAILURE in 11m 33s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/13/upstream-check/cinder-cDOT-FCP/d75b557/ : FAILURE in 13m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"0670e428edff4df9da1afe1deb53ebee30ce0c02","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 22:15:13.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8f62f20/8f62f20.html : SUCCESS in 2h 27m 33s","accounts_in_message":[],"_revision_number":13},{"id":"ceae599e349b6039520229c0b259a2eda6ceb5c3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-20 23:46:52.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/13/check/DellEMC_PowerStore_iSCSI/dc76a31/DellEMC_PowerStore_iSCSI/111 : FAILURE in 49m 51s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/13/check/DellEMC_PowerStore_FC/ee0c927/DellEMC_PowerStore_FC/112 : SUCCESS in 48m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"7bc643e809e7201f4b78744c62c77ed11f0d9619","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 04:26:00.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_nfs : SUCCESS in 4h 43m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_iscsi : SUCCESS in 4h 47m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"eebdb0fa25e89dc75c01bf1189ee097524964170","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-21 05:40:46.000000000","message":"Patch Set 13:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-13/Seagate-iSCSI : SUCCESS in 45:43.46\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-13/Seagate-FC : SUCCESS in 39:36.23","accounts_in_message":[],"_revision_number":13},{"id":"4d8053cffbe4292113162e805c76d165bbc66325","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-21 07:12:17.000000000","message":"Patch Set 13:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-13/Lenovo-iSCSI : SUCCESS in 51:24.59\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-13/Lenovo-FC : SUCCESS in 46:54.57","accounts_in_message":[],"_revision_number":13},{"id":"932088c92d4e0137453deb79a58895c04099f541","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 07:54:21.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/13/check/DellEMC_XtremIO_FC/8e6b391/DellEMC_XtremIO_FC/186 : FAILURE in 1h 05m 39s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/13/check/DellEMC_XtremIO_iSCSI/9a0fb61/DellEMC_XtremIO_iSCSI/193 : SUCCESS in 1h 48m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":13},{"id":"134dafe60c058c4151697230482b6f2ef29a6aaa","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-21 08:45:16.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/13/check/DellEMC_SC_FC/d0b7e35/DellEMC_SC_FC/127 : FAILURE in 1h 03m 28s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/13/check/DellEMC_SC_iSCSI/7efcf21/DellEMC_SC_iSCSI/124 : SUCCESS in 2h 16m 05s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":13},{"id":"82389a670cd0b6cae4499ef60091a8e314f96c01","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-21 08:55:29.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/13/check/DellEMC_Unity_FC/922996a/DellEMC_Unity_FC/199 : SUCCESS in 1h 16m 50s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/13/check/DellEMC_Unity_iSCSI/c765220/DellEMC_Unity_iSCSI/185 : SUCCESS in 1h 46m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"8ab26cf5268105e29ec6ab75aa775c6fa0d7e06b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-21 09:58:43.000000000","message":"Patch Set 13:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-13/HPMSA-iSCSI : FAILURE in 1:26:19\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-13/HPMSA-FC : SUCCESS in 33:47.44","accounts_in_message":[],"_revision_number":13},{"id":"b13a72577323ac70573ac8ad310249e22288c5f7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 17:39:06.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/13/check/DellEMC_VNX_iSCSI/96e8c27/DellEMC_VNX_iSCSI/128 : FAILURE in 3h 54m 06s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/13/check/DellEMC_VNX_FC/8ef3451/DellEMC_VNX_FC/133 : SUCCESS in 1h 32m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":13},{"id":"0dd030f098c211de3c67009fd807994395ffc3da","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-22 06:36:49.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/23/784623/13/check/DellEMC_PowerMAX_FC/20ebed1/DellEMC_PowerMAX_FC/89 : FAILURE in 4h 39m 28s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/23/784623/13/check/DellEMC_PowerMAX_iSCSI/3435a83/DellEMC_PowerMAX_iSCSI/79 : SUCCESS in 4h 33m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":13},{"id":"defad4838f771e59105d98df967d1da64fcc326e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-24 02:16:46.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/13/20210524083910/NEC-FC : SUCCESS in 2h 36m 51s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/13/20210524083910/NEC-ISCSI : SUCCESS in 2h 37m 33s","accounts_in_message":[],"_revision_number":13},{"id":"ab07aae173ba54ce55ba9981b089e74673b6a418","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-28 04:10:54.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_nfs : SUCCESS in 6h 20m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_iscsi : SUCCESS in 5h 57m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"88401f8da6a73863535f0a10f25962cb1b747111","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-28 21:02:50.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_nfs : FAILURE in 53m 41s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_iscsi : SUCCESS in 4h 48m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"4b7903c07c55015ea329859e07049737a068757a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-28 21:39:33.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_nfs : SUCCESS in 4h 54m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_iscsi : SUCCESS in 5h 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"4e6174f676ae746e2b68f0205a16d19c450d6c86","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-30 13:33:50.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_nfs : FAILURE in 6h 32m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_iscsi : FAILURE in 6h 33m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"19f9eefd7c1decbd2bad5b6edad1c007d3d8d263","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-31 08:14:42.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_nfs : SUCCESS in 5h 1m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-13/ns5_iscsi : SUCCESS in 5h 7m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"c645d2b03951d9eb8b1cf074340b19866af75249","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-07-13 05:59:41.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"aa14763b5482a396efe893deac9e04b445a82cf2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-20 18:13:30.000000000","message":"Topic bug/1922408 removed","accounts_in_message":[],"_revision_number":13},{"id":"b68fb38e869210149a26df438e1d7ca68c82e66f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-20 18:13:33.000000000","message":"Topic set to rbd-encrypt-issues","accounts_in_message":[],"_revision_number":13},{"id":"054d02461915af7b7c96633ef8ae77015f55736d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-07-21 01:11:21.000000000","message":"Patch Set 13: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":13},{"id":"ad915ac8579c6b5ffbe09399d0f76b57247e3b9f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-27 19:31:50.000000000","message":"Patch Set 13: Code-Review+1\n\nThis logic is sound to me.  The case where encryption is not used, the volume and snapshot are the same size and no resize is needed.  In the case where the caller wants a larger volume and still no encryption, _resize() is called as it should.  And in the case where encryption is being used, we calculate a correct size to avoid truncation.  Looks good to me.","accounts_in_message":[],"_revision_number":13},{"id":"25c0ab3d7aeb07646e6b25d460b3ea56677d58ba","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-27 19:55:06.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9480fe892d8f5fbbc26c30438d6339c4cc7180c1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-07-28 14:46:39.000000000","message":"Patch Set 13: Code-Review+2\n\nI think this makes sense to me.  Thank you!","accounts_in_message":[],"_revision_number":13},{"id":"a26f26e220e2a7fe74ef9865172fb45cace1a5bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-28 17:45:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d390e732a0096b15d1eab030b0d1776dc288fa9f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-28 17:45:59.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"d61ba3797ee2a1bcd94e03f2254e87fef7eafd32","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 17:52:25.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"9fb1a32a14eac41fdc59219d39450c2c9905f24f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-28 18:01:27.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f5afc3d03f39930808a20afa8227e5fe10b7205d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:13:21.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2ef26ce90474ea0b0146729dca325d1 : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a46e433e4804d44bf6cf8f592fe29b0 : SUCCESS in 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d7c79056677a219e559bb35b7720699672c02a1e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-28 18:36:22.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/14/iscsi/ : SUCCESS in 50m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/14/smb/ : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":14},{"id":"f9d6c12520d9110a56017bb562ef2d401e114e62","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-28 18:51:06.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/784623/14/ : FAILURE in 56m 36s","accounts_in_message":[],"_revision_number":14},{"id":"f0c3c9e17bfb339d1ec995967c9a6ec7993cd77f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 18:52:29.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/14/check/iscsi-driver-primera/6267038/6267038.html : FAILURE in 1h 05m 28s","accounts_in_message":[],"_revision_number":14},{"id":"a38e5815b23251507680a14bc396a9670612437a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-28 18:54:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/14/check/dsvm-tempest-cisco-zonemanager-job/fc3dd9b/ : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":14},{"id":"8db5fe2161d03c7b02cbc729c44ade5d88724002","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 19:08:50.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/43c5d5c/43c5d5c.html : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":14},{"id":"21df689e4ce7512261ddf39cbb2b541da7bb06be","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-28 20:44:02.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/14/check/DellEMC_Unity_FC/17cf70b/DellEMC_Unity_FC/522 : SUCCESS in 1h 02m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/14/check/DellEMC_Unity_iSCSI/c3c1e37/DellEMC_Unity_iSCSI/493 : SUCCESS in 1h 43m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"50ccf4f8b23c421e9a5099fc30e36b938b19acfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-28 21:00:29.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"96d5b32e279f4d834bf665cae6705b10930201e1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-28 21:01:15.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"a455788cdf4e1f83274339067b5aaa85004520a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 21:22:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecfe890d7adc42acaa71f17bee7ed2fe : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f9e5b9dca6a4323aa3c78edc1380387 : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"10386f446a6883a187b6e092b7799c1854f05400","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-28 21:54:34.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/15/iscsi/ : SUCCESS in 53m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/15/smb/ : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":15},{"id":"e4eeb9a12bdef46500fcec849de85af1c9e443ad","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-28 21:56:29.000000000","message":"Patch Set 14:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-14/ns5_nfs : FAILURE in 3h 59m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-14/ns5_iscsi : FAILURE in 4h 2m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"f883597209bcb6f0e31dbfba1d1d48fd2cb20b51","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-28 22:03:56.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/784623/15/ : FAILURE in 58m 41s","accounts_in_message":[],"_revision_number":15},{"id":"00ed15c76b49232ecb3904ca680da176d930dcd5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 22:05:03.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/15/check/iscsi-driver-primera/63d5787/63d5787.html : FAILURE in 1h 03m 06s","accounts_in_message":[],"_revision_number":15},{"id":"329b6e961c7a22e78798cd216074772a65e56ba4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-28 22:09:12.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/15/check/dsvm-tempest-cisco-zonemanager-job/e282a99/ : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":15},{"id":"37e68dcf1ce087272a4199bd1c835831e31f97a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 22:27:38.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/80450c2/80450c2.html : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":15},{"id":"be8ae4f25af1f2a28957a7b785c8699038037e60","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-28 22:51:38.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/15/check/DellEMC_Unity_FC/638d2a8/DellEMC_Unity_FC/523 : FAILURE in 1h 16m 35s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/15/check/DellEMC_Unity_iSCSI/0291c3a/DellEMC_Unity_iSCSI/494 : SUCCESS in 1h 50m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"06ea0b8d868c37dc095cbe61f5e32d3fc057e8bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 23:16:20.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7dbbcd5726f458cb0712b966c80e5bf : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/744ca3ab01244956b3ea0e7ea2a9c6be : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d366146873ef4f47bc461f06065eb059 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8186636520d444bb89cbd1bf247ac00d : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18ea91073414d15a5b611aeeed7fb72 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/98d875aa18bb4365b0d98ee77e0d871d : SUCCESS in 1h 08m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76f7ee5f1dad4b51aca69a3486adc667 : SUCCESS in 56m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffd44296fddb42afb1bf7cc469295b57 : SUCCESS in 5m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95b770d36b8340b998ee55e084c46d70 : SUCCESS in 13m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb0818f9a4db4e2898dddc843313cab3 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/490d132188f345cba2e92e1b8bccb62a : SUCCESS in 6m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9843906d020c417bb10b533b38cbcd8e : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/db9e16a2deb04a6a80fd0328482d1058 : SUCCESS in 33m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54217288fede4a8098fa24ecc19119a0 : FAILURE in 8m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b3936d82ff64ba68517de3f152355b1 : SUCCESS in 1h 12m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b6ef3684aecd4e36ba32e434765e6f48 : SUCCESS in 1h 22m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5ba335823b9c4b8388073070e3417656 : SUCCESS in 1h 32m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3244efa8271c4c2589db7f91eeb9ca39 : SUCCESS in 1h 20m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13074290ef5647419d4fc0c4500aedb7 : SUCCESS in 1h 02m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd34aa9d22364e57b0590e659f77d095 : SUCCESS in 1h 02m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1c15a154d5440ea8c4bc0d0f6e611f3 : SUCCESS in 2h 08m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a35f9a7b4ca44a0b28de0f853b18de5 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":15},{"id":"eaff9a6b00914d2cdd81d16695183c9122a7fc0e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-29 00:09:07.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/15/check/DellEMC_PowerStore_iSCSI/59dcd26/DellEMC_PowerStore_iSCSI/252 : FAILURE in 52m 21s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/15/check/DellEMC_PowerStore_FC/2efccc7/DellEMC_PowerStore_FC/276 : SUCCESS in 44m 10s\n- DellEMC_PowerStore_NFS http://publiclogs.emc.com/23/784623/15/check/DellEMC_PowerStore_NFS/25fab7f/DellEMC_PowerStore_NFS/34 : FAILURE in 29m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":15},{"id":"a08a3c44710ce5131effddf65ec79130005e63ec","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-29 00:13:21.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/15/upstream-check/cinder-cDOT-iSCSI/47344bc/ : SUCCESS in 36m 09s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/15/upstream-check/cinder-cDOT-NFS/d9c4326/ : SUCCESS in 37m 58s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/15/upstream-check/cinder-cDOT-FCP/929fb87/ : FAILURE in 55m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"3d96b4abd81833154ace5c73baad9ff7b8814c86","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-29 00:27:59.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/15/check/DellEMC_SC_FC/16dff48/DellEMC_SC_FC/195 : SUCCESS in 50m 41s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/15/check/DellEMC_SC_iSCSI/674250f/DellEMC_SC_iSCSI/191 : SUCCESS in 1h 11m 15s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"a38065662f6e10052e9a5e4039708a2a9a3209e2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 00:56:56.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/15/check/DellEMC_PowerFlex/291037c/DellEMC_PowerFlex/312 : SUCCESS in 1h 40m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"10c6f1f2982c421b9fbce5e56c2eb07dceb6fe0b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-29 00:57:44.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/15/check/DellEMC_Unity_FC/be427be/DellEMC_Unity_FC/524 : SUCCESS in 1h 15m 30s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/15/check/DellEMC_Unity_iSCSI/adcfad6/DellEMC_Unity_iSCSI/495 : SUCCESS in 1h 41m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"598aae0285dc7163caba31f0b320d3e5883e4cbc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-29 01:00:41.000000000","message":"Patch Set 15:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-15/ns5_nfs : FAILURE in 3h 58m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-15/ns5_iscsi : FAILURE in 3h 59m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"d86afc43de44ed9d2a2fec6de9c1cb679f1e1aca","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-29 01:01:20.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/15/upstream-solidfire-check/cinder-solidfire-iSCSI/99e4e12/ : FAILURE in 1h 43m 33s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"ea7ceb2dcaad06eeabd282e560186a6135684380","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-29 01:04:09.000000000","message":"Patch Set 15:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-15/Seagate-iSCSI : FAILURE in 39:52.74\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-15/Seagate-FC : FAILURE in 37:48.58","accounts_in_message":[],"_revision_number":15},{"id":"f7aa84be7d548304a6e839b754cff92cff821f3e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-29 01:16:48.000000000","message":"Patch Set 15:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-15/HPMSA-iSCSI : FAILURE in 55:40.03\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-15/HPMSA-FC : FAILURE in 32:15.70","accounts_in_message":[],"_revision_number":15},{"id":"3582b96bc4456e856b39af6f7a89d0ac79d47f98","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-29 01:19:34.000000000","message":"Patch Set 15:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-15/Lenovo-iSCSI : FAILURE in 46:53.81\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-15/Lenovo-FC : FAILURE in 46:03.56","accounts_in_message":[],"_revision_number":15},{"id":"9e98a8d7aaef7701ec465e7c6dc9265afc33f203","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-29 01:50:10.000000000","message":"Patch Set 15:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-15/PVME-iSCSI : FAILURE in 1:15:20\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-15/PVME-FC : FAILURE in 46:09.42","accounts_in_message":[],"_revision_number":15},{"id":"3b52bff843dc2ae4767d9ace5d0f4f81cb871d90","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 05:51:55.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/15/check/DellEMC_VNX_iSCSI/2dc29eb/DellEMC_VNX_iSCSI/274 : SUCCESS in 3h 05m 56s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/15/check/DellEMC_VNX_FC/e94178f/DellEMC_VNX_FC/292 : SUCCESS in 1h 43m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"278fcad4523a9c93fdcb7b2baf88908ad3ded0a2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-07-29 14:10:56.000000000","message":"Patch Set 15: Code-Review-1\n\n(4 comments)\n\nCode LGTM, i would like to see the test (mentioned inline) added since it\u0027s also a case we are fixing here.","accounts_in_message":[],"_revision_number":15},{"id":"9a096ad40bb86c4056c30a8d678e97fbdad22e28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-29 16:33:53.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"90f9a6b71cf6aa5fee3441ac2eaaa9739b7f2fb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 16:54:13.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e936a960183a4634a1a880dbbd419376 : SUCCESS in 16m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e8cb7ae0deb417d9634f1cd97827abc : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"707308e3de50929d133cab6e565cfc80c87e5aa8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-07-29 17:13:32.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)\n\nThanks for addressing my comments, LGTM.","accounts_in_message":[],"_revision_number":16},{"id":"2bfc86e0d6c57ef20c73975d3aa60d6d40e87adc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-29 17:26:36.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/784623/16/iscsi/ : SUCCESS in 51m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/784623/16/smb/ : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":16},{"id":"c23c34a4b60cbdaf298a3db3c4a218cd88caf65f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-29 17:36:31.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/784623/16/check/iscsi-driver-primera/8b55562/8b55562.html : FAILURE in 1h 01m 07s","accounts_in_message":[],"_revision_number":16},{"id":"9a9d759080619088ac29a2c2dd7f138b6b4ebf9b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-29 17:52:31.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/784623/16/ : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":16},{"id":"f8e6c68621600682f8784d4c47e7b72ef487ce07","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-29 18:08:55.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/784623/16/check/dsvm-tempest-cisco-zonemanager-job/df62c6a/ : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":16},{"id":"44b10e3cd615fb15b3e183de76f9c5125f42107a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:29:14.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd0f528434846fea23d4463f1606a08 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8731ce0e7ef54f44b6adfebabf919918 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c60c1e8080b4ad4b3928283ad49fa09 : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e6e33284b094e52883fc168e9a91ebd : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6731f177770d498bb77422e99789f7d7 : SUCCESS in 15m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9bb44c0d10e8422c867d82249a2c4cb3 : SUCCESS in 59m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67dd494eb0444eddb88a9fb67472f897 : SUCCESS in 1h 31m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa35382f784645e0864449f4977649d4 : SUCCESS in 6m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1a2d14f3a8d94b2682eb1250052cdd9d : SUCCESS in 13m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67fe08ec59904c6295d365ee1c1ca259 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0bf3971a070a40f5b3cd7652edadb96c : SUCCESS in 6m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a4e21aa54ace4cb890e4aaac79d6f12c : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c81c6e5067e04cd180d7736ac043d1cc : SUCCESS in 31m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a735f3921c3446689210db9da104edef : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0306ea1a71c549a6b4406cb3baa34c12 : SUCCESS in 1h 04m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ca3c7e7ce2ff43d385d6cc272bd0ccba : SUCCESS in 1h 26m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1abd400f427a4b7ba3f3ac49c349002a : SUCCESS in 1h 22m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aef899c36a5a4388871fd538c840a1c0 : SUCCESS in 1h 02m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b0f9eaba54f401ba424f6e45fe97bf5 : SUCCESS in 55m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f7ed9d7728794ed78f97133d4c3d7ada : SUCCESS in 1h 10m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd492b85260f4c96966019492e4383fe : SUCCESS in 1h 52m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1685874ede234f45a7c4e3ebc44445a1 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":16},{"id":"493dc3ece8531a3545c938d05d44c438e5cccca5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-29 18:30:30.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/16/check/tempest-dsvm-full-focal-py3/2af96f3/ : FAILURE in 1h 51m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/23/784623/16/check/devstack-plugin-ceph-dsvm-focal-py3/0792d78/ : FAILURE in 43m 05s","accounts_in_message":[],"_revision_number":16},{"id":"747d75eee9ec9cf013025600ac5ce171edf86833","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-29 18:31:29.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"2467fd412445a415b0396b92f8d4f65fd8b7c51d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-29 18:41:20.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/23/784623/16/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/576e773/ : FAILURE in 2h 05m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/23/784623/16/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ac52d77/ : FAILURE in 2h 05m 59s","accounts_in_message":[],"_revision_number":16},{"id":"fc3ed8bbafaae924a213fd91eba0e7d1bde2001b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-29 19:35:36.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/784623/16/upstream-check/cinder-cDOT-iSCSI/feda868/ : SUCCESS in 37m 51s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/784623/16/upstream-check/cinder-cDOT-NFS/fd3f184/ : SUCCESS in 37m 45s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/784623/16/upstream-check/cinder-cDOT-FCP/e646b9d/ : FAILURE in 1h 05m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"de588de983c233c7db324b0e39b3354c49adbde9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-29 19:48:58.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/784623/16/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b5327c2/b5327c2.html : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":16},{"id":"3fe4c95d1f1ea5a643db9e2801350ca1ee6299b2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-29 19:57:03.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/23/784623/16/check/DellEMC_PowerStore_iSCSI/05c0f19/DellEMC_PowerStore_iSCSI/257 : FAILURE in 58m 35s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/23/784623/16/check/DellEMC_PowerStore_FC/08dda88/DellEMC_PowerStore_FC/281 : SUCCESS in 50m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":16},{"id":"c8b24637d0cc7f18607f1ef3adccc1cb3f1021e9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-29 20:13:03.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/784623/16/upstream-solidfire-check/cinder-solidfire-iSCSI/13c9d3d/ : FAILURE in 1h 42m 40s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":16},{"id":"ad938404dd5c1a5418ad9041481f4f684cb0be39","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-29 21:00:42.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/23/784623/16/check/DellEMC_SC_FC/1eb03d4/DellEMC_SC_FC/199 : SUCCESS in 48m 21s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/23/784623/16/check/DellEMC_SC_iSCSI/5030eb0/DellEMC_SC_iSCSI/195 : SUCCESS in 1h 17m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":16},{"id":"13b699f2e397410634181b3ee11106e0015844b0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-29 21:08:29.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/23/784623/16/check/DellEMC_Unity_FC/491cb83/DellEMC_Unity_FC/530 : SUCCESS in 1h 18m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/23/784623/16/check/DellEMC_Unity_iSCSI/7e54ec6/DellEMC_Unity_iSCSI/501 : SUCCESS in 1h 59m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"790a547e00c66e047134da4e5392caadc514a336","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-29 21:16:13.000000000","message":"Patch Set 16:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-16/ns5_nfs : FAILURE in 4h 21m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-784623-16/ns5_iscsi : FAILURE in 4h 26m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"1d56382cf1e3334c2cd854cecacef9fe6762e426","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 22:13:07.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/23/784623/16/check/DellEMC_PowerFlex/1722020/DellEMC_PowerFlex/316 : FAILURE in 1h 44m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"a396ee70f34643e4e7f8fd6cac73ceed85a4645f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 22:31:36.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/23/784623/16/check/DellEMC_XtremIO_FC/c76fe5d/DellEMC_XtremIO_FC/564 : FAILURE in 1h 23m 16s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/23/784623/16/check/DellEMC_XtremIO_iSCSI/c39e087/DellEMC_XtremIO_iSCSI/551 : SUCCESS in 2h 03m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":16},{"id":"6ff77b38ef8217fb48e1028f0d560275abcddceb","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-29 22:44:13.000000000","message":"Patch Set 16:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/784623-16/Seagate-iSCSI : FAILURE in 39:50.29\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/784623-16/Seagate-FC : FAILURE in 37:46.83","accounts_in_message":[],"_revision_number":16},{"id":"9b98f812fbc3448a59fbbf2158cffe180ca4fdc8","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-29 23:21:02.000000000","message":"Patch Set 16:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/784623-16/Lenovo-iSCSI : FAILURE in 45:55.75\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/784623-16/Lenovo-FC : FAILURE in 44:08.86","accounts_in_message":[],"_revision_number":16},{"id":"50eec540e9a7ccd83247560ed4b51284c91b2bc0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-29 23:38:29.000000000","message":"Patch Set 16:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/784623-16/HPMSA-iSCSI : FAILURE in 57:07.14\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/784623-16/HPMSA-FC : FAILURE in 32:02.64","accounts_in_message":[],"_revision_number":16},{"id":"5fecebfd3421850258e9fd56f535bba5e1f02462","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-30 01:52:15.000000000","message":"Patch Set 16:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/784623-16/PVME-iSCSI : FAILURE in 1:34:55\n* PVME-FC http://os-logs.tristero.net/dell-ci/784623-16/PVME-FC : FAILURE in 46:11.16","accounts_in_message":[],"_revision_number":16},{"id":"eb6f381fb3da9c2e373f3612c74b723e93f79c16","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-30 04:40:02.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/23/784623/16/check/DellEMC_VNX_iSCSI/ce66ee3/DellEMC_VNX_iSCSI/278 : SUCCESS in 3h 35m 58s\n- DellEMC_VNX_FC http://publiclogs.emc.com/23/784623/16/check/DellEMC_VNX_FC/69c6657/DellEMC_VNX_FC/296 : SUCCESS in 1h 28m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"42b037a392f2089046392414296192138752da91","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-07-30 17:25:27.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nThis looks good to me and is consistent with fixes made elsewhere.","accounts_in_message":[],"_revision_number":16},{"id":"93e338ed9772f844d5e1c5d8c9a617b34d8c4424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:26:03.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"618fc6cf50c0191389c27e52498ed2f8991f8e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 19:05:37.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74aa2dd90e8e47eba31ed29b391faafd : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98aa8540eea9453dbcab469b923090a5 : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26b7ab1851554f50bf59a8cf8975bf43 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cf1df562044c7c8d16b634c680e223 : SUCCESS in 11m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/bf20fd7eadf147e1a1b3d8d045d732d7 : SUCCESS in 1h 04m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3979044142b4ccab2f850770dec961e : SUCCESS in 1h 03m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e82ee4481d546269b4312a72ded3cff : SUCCESS in 6m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3046ebf43ac245018914a8c2526d9295 : SUCCESS in 1h 14m 28s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e6e93862ef974c609aa06ed57d976060 : SUCCESS in 1h 12m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf8cb7be4071420ab99000835ec6bc29 : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":16},{"id":"5671a834a84b7187bd14ed3ae01da2c6405b25c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 19:05:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"7824055da2322bd8728898aef2aab571e29203a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 19:07:45.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86037be9b2948fba9d2a00178593dc9 : SUCCESS in 1m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9c85287bc95444faa95fa92b6b18721 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":16},{"id":"ff306934916cdbc8f8d15066a05e179a677bc605","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-03 04:13:59.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/23/784623/16/check/dsvm-tempest-TOYOU-ACS5000-driver/fc70c55 : FAILURE in 24m 15s","accounts_in_message":[],"_revision_number":16},{"id":"6118c65bf34565cc86241886de2ffce3dde31345","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-08-10 02:07:02.000000000","message":"Patch Set 15:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/15/20210810103558/NEC-ISCSI : FAILURE in 25m 19s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/784623/15/20210810103558/NEC-FC : FAILURE in 30m 59s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":16,"current_revision":"cf1c5252965e731d77af503529ca42e269c305c6","revisions":{"bf51a664569ca6fb42d2080e9b60aeb166125450":{"kind":"REWORK","_number":1,"created":"2021-04-03 04:38:13.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/1"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:48.000000000","tz":480},"subject":"Fixed rbd driver create encryption volume from snapshot causing encryption header loss","message":"Fixed rbd driver create encryption volume from snapshot causing encryption header loss\n\nIn rbd driver, encryption volume created from snapshot that source\nvolume is also encrypted. if the size of new volume not greater than\nthe size of source volume, will skip resize new volume to prevent\nthe loss of the encryption header.\n\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf51a664569ca6fb42d2080e9b60aeb166125450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf51a664569ca6fb42d2080e9b60aeb166125450"}]},"branch":"refs/heads/master"},"96f89306548d9d5e246d58e20aa6f94c358d6541":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-03 04:39:48.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/2"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:39:48.000000000","tz":0},"subject":"Fixed rbd driver create encryption volume from snapshot causing encryption header loss","message":"Fixed rbd driver create encryption volume from snapshot causing encryption header loss\n\nIn rbd driver, encryption volume created from snapshot that source\nvolume is also encrypted. if the size of new volume not greater than\nthe size of source volume, will skip resize new volume to prevent\nthe loss of the encryption header.\n\nCloses-Bug:#1922408\n\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96f89306548d9d5e246d58e20aa6f94c358d6541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96f89306548d9d5e246d58e20aa6f94c358d6541"}]},"branch":"refs/heads/master","description":"Edit commit message"},"85b50d3bee90797086bb78098bcfe44cbb974e05":{"kind":"REWORK","_number":3,"created":"2021-04-07 06:35:05.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/3"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-07 06:34:26.000000000","tz":480},"subject":"Fixed rbd driver create encryption volume from snapshot causing encryption header loss","message":"Fixed rbd driver create encryption volume from snapshot causing encryption header loss\n\nIn rbd driver, encryption volume created from snapshot that source\nvolume is also encrypted. if the size of new volume not greater than\nthe size of source volume, will skip resize new volume to prevent\nthe loss of the encryption header.\n\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85b50d3bee90797086bb78098bcfe44cbb974e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85b50d3bee90797086bb78098bcfe44cbb974e05"}]},"branch":"refs/heads/master"},"f04faf3e06ad700ec6f19e8783e3585fde59b3f8":{"kind":"REWORK","_number":4,"created":"2021-04-08 19:02:11.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/4"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-04-08 18:59:25.000000000","tz":0},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f04faf3e06ad700ec6f19e8783e3585fde59b3f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f04faf3e06ad700ec6f19e8783e3585fde59b3f8"}]},"branch":"refs/heads/master"},"316000cfd4fba1da33fb71401aff7b1aad702eea":{"kind":"REWORK","_number":5,"created":"2021-04-09 00:18:30.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/5"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-04-09 00:18:04.000000000","tz":0},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/316000cfd4fba1da33fb71401aff7b1aad702eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/316000cfd4fba1da33fb71401aff7b1aad702eea"}]},"branch":"refs/heads/master"},"bf75d2523fc1e0b2e8308f7674326fc3501f204e":{"kind":"REWORK","_number":6,"created":"2021-04-09 01:09:38.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/6"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-09 01:08:45.000000000","tz":480},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf75d2523fc1e0b2e8308f7674326fc3501f204e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf75d2523fc1e0b2e8308f7674326fc3501f204e"}]},"branch":"refs/heads/master"},"64b8a083f3e2579a30e2bb08384296449aa4a99e":{"kind":"REWORK","_number":7,"created":"2021-04-19 20:09:52.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/7"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-04-19 20:09:40.000000000","tz":0},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b8a083f3e2579a30e2bb08384296449aa4a99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b8a083f3e2579a30e2bb08384296449aa4a99e"}]},"branch":"refs/heads/master"},"1c2912d79b5fd2f66461e573315dde94bd27dec6":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-05-05 23:57:54.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/8"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-05-05 23:57:20.000000000","tz":0},"subject":"[rbd] Fix create encrypted volume from snapshot","message":"[rbd] Fix create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot Cinder\nwill silently truncate the data in the new volume. As a\nresult the volume can\u0027t be accessed after attachingto an\ninstance.\n\nIn particular, RBD stores the real size of an encrypted\nvolume and may differ from what openstack assumes (the\nsize may vary depending on the encryption parameters).\n\nIn order to fix this this patch only forces resizing\nwhen the destination volume size is bigger than the\nsourcevolume size. In case the destination volume size is\nbigger than the source volumes we should calculate the\nnew size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c2912d79b5fd2f66461e573315dde94bd27dec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c2912d79b5fd2f66461e573315dde94bd27dec6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5ec2996a06034eb5ad0d91f082eed97f8434acf4":{"kind":"REWORK","_number":9,"created":"2021-05-18 01:02:07.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/9"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-05-18 01:00:47.000000000","tz":480},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ec2996a06034eb5ad0d91f082eed97f8434acf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ec2996a06034eb5ad0d91f082eed97f8434acf4"}]},"branch":"refs/heads/master"},"64d23be67f264d6381bfe1d72a8120e3b5ecea1b":{"kind":"REWORK","_number":10,"created":"2021-05-18 03:21:03.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/10"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-05-18 03:20:35.000000000","tz":480},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64d23be67f264d6381bfe1d72a8120e3b5ecea1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64d23be67f264d6381bfe1d72a8120e3b5ecea1b"}]},"branch":"refs/heads/master"},"4ebe2fd38be9274ea3577b74b0c4ed6c6ae3c58a":{"kind":"REWORK","_number":11,"created":"2021-05-19 01:44:29.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/23/784623/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/11"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-05-19 01:44:00.000000000","tz":480},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ebe2fd38be9274ea3577b74b0c4ed6c6ae3c58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ebe2fd38be9274ea3577b74b0c4ed6c6ae3c58a"}]},"branch":"refs/heads/master"},"dde5a8fa46f97aeba942e3b5cae5232f9799aa24":{"kind":"REWORK","_number":12,"created":"2021-05-19 20:26:36.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/12"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-05-19 20:25:59.000000000","tz":0},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dde5a8fa46f97aeba942e3b5cae5232f9799aa24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dde5a8fa46f97aeba942e3b5cae5232f9799aa24"}]},"branch":"refs/heads/master"},"c3364909470577112084df4bab6662d066e24edd":{"kind":"REWORK","_number":13,"created":"2021-05-20 19:44:17.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/13"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-05-20 19:44:08.000000000","tz":0},"subject":"Fix RBD create encrypted volume from snapshot","message":"Fix RBD create encrypted volume from snapshot\n\nWhen creating an encrypted volume from a snapshot of an\nencrypted volume, if the amount of data in the original\nvolume at the time the snapshot was created is very close\nto the gibibyte boundary given by the volume\u0027s size, it\nis possible for the data in the new volume to be silently\ntruncated.\n\nIn order to fix this only force resize when the destination\nvolume size is bigger than the source volumes and calculate\nthe new size based on the difference and not in round\nnumbers.\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3364909470577112084df4bab6662d066e24edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3364909470577112084df4bab6662d066e24edd"}]},"branch":"refs/heads/master"},"7e0f4aacdaa6a91ebece99b1f0328fb3938dbc7d":{"kind":"REWORK","_number":14,"created":"2021-07-28 17:45:00.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/14"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-07-28 17:44:31.000000000","tz":0},"subject":"[rbd] Fix create encrypted volume from snapshot","message":"[rbd] Fix create encrypted volume from snapshot\n\nUsually the source volume would be the same size or smaller\nthan the destination volume and they must share the same\nvolume-type. Currently creating an encrypted volume from a\nsnapshot of an encrypted volume truncates the data in the\nnew volume.\n\nIn order to fix this the RBD workflow would be something\nlike this:\nA source luks volume would be 1026M, we write some data\nand create a snap from it. We like to create a new luks\nvolume from a snapshot so the create_volume_from_snapshot()\nmethod performs a RBD clone first and then a resize if needed.\n\nIn addition the _clone() method creates a clone\n(copy-on-write child) of the parent snapshot. Object size\nwill be identical to that of the parent image unless specified\n(we don\u0027t in cinder) so size will be the same as the parent\nsnapshot.\n\nIf the desired size of the destination luks volume is 1G the\ncreate_volume_from_snapshot() won\u0027t perform any resize and\nwill be 1026M as the parent. This solves bug #1922408 because\nwe don\u0027t force it to resize and because of that we don\u0027t\ntruncate the data anymore.\n\nThe second case scenario is when we would like to increase\nthe size of the destination volume. As far as I can tell this\nwon\u0027t face the encryption header problem but we still need to\ncalculate the difference size to provide the size that the\nuser is expecting.\n\nThat\u0027s why the fix proposed calculate the new_size based on:\nsize difference \u003d desired size - size of source volume\nnew size \u003d current size + size difference\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e0f4aacdaa6a91ebece99b1f0328fb3938dbc7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e0f4aacdaa6a91ebece99b1f0328fb3938dbc7d"}]},"branch":"refs/heads/master"},"3bf6961e54469c2233121c3587c66d76c9a47ee3":{"kind":"REWORK","_number":15,"created":"2021-07-28 21:00:29.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/15"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-07-28 21:00:03.000000000","tz":0},"subject":"[rbd] Fix create encrypted volume from snapshot","message":"[rbd] Fix create encrypted volume from snapshot\n\nUsually the source volume would be the same size or smaller\nthan the destination volume and they must share the same\nvolume-type. Currently creating an encrypted volume from a\nsnapshot of an encrypted volume truncates the data in the\nnew volume.\n\nIn order to fix this the RBD workflow would be something\nlike this:\nA source luks volume would be 1026M, we write some data\nand create a snap from it. We like to create a new luks\nvolume from a snapshot so the create_volume_from_snapshot()\nmethod performs a RBD clone first and then a resize if needed.\n\nIn addition the _clone() method creates a clone\n(copy-on-write child) of the parent snapshot. Object size\nwill be identical to that of the parent image unless specified\n(we don\u0027t in cinder) so size will be the same as the parent\nsnapshot.\n\nIf the desired size of the destination luks volume is 1G the\ncreate_volume_from_snapshot() won\u0027t perform any resize and\nwill be 1026M as the parent. This solves bug #1922408 because\nwe don\u0027t force it to resize and because of that we don\u0027t\ntruncate the data anymore.\n\nThe second case scenario is when we would like to increase\nthe size of the destination volume. As far as I can tell this\nwon\u0027t face the encryption header problem but we still need to\ncalculate the difference size to provide the size that the\nuser is expecting.\n\nThat\u0027s why the fix proposed calculate the new_size based on:\nsize difference \u003d desired size - size of source volume\nnew size \u003d current size + size difference\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bf6961e54469c2233121c3587c66d76c9a47ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bf6961e54469c2233121c3587c66d76c9a47ee3"}]},"branch":"refs/heads/master"},"cf1c5252965e731d77af503529ca42e269c305c6":{"kind":"REWORK","_number":16,"created":"2021-07-29 16:33:53.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/23/784623/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/784623/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/784623/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/784623/16"}}},"commit":{"parents":[{"commit":"393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18","subject":"Merge \"Quota: Fix multiple race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/393c2e4ad90c05ebf28cc3a2c65811d7e1e0bc18"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-04-03 04:36:14.000000000","tz":480},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-07-29 16:31:21.000000000","tz":0},"subject":"[rbd] Fix create encrypted volume from snapshot","message":"[rbd] Fix create encrypted volume from snapshot\n\nUsually the source volume would be the same size or smaller\nthan the destination volume and they must share the same\nvolume-type. In particular, when the destination volume is\nsame size as the source volume, creating an encrypted volume\nfrom a snapshot of an encrypted volume truncates the data in\nthe new volume.\n\nIn order to fix this the RBD workflow would be something\nlike this:\nA source luks volume would be 1026M, we write some data\nand create a snap from it. We like to create a new luks\nvolume from a snapshot so the create_volume_from_snapshot()\nmethod performs a RBD clone first and then a resize if needed.\n\nIn addition the _clone() method creates a clone\n(copy-on-write child) of the parent snapshot. Object size\nwill be identical to that of the parent image unless specified\n(we don\u0027t in cinder) so size will be the same as the parent\nsnapshot.\n\nIf the desired size of the destination luks volume is 1G the\ncreate_volume_from_snapshot() won\u0027t perform any resize and\nwill be 1026M as the parent. This solves bug #1922408 because\nwe don\u0027t force it to resize and because of that we don\u0027t\ntruncate the data anymore.\n\nThe second case scenario is when we would like to increase\nthe size of the destination volume. As far as I can tell this\nwon\u0027t face the encryption header problem but we still need to\ncalculate the difference size to provide the size that the\nuser is expecting.\n\nThat\u0027s why the fix proposed calculate the new_size based on:\nsize difference \u003d desired size - size of source volume\nnew size \u003d current size + size difference\n\nCloses-Bug: #1922408\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\nChange-Id: I220b5e3b01d115262a8b1dd45758f0531aea0edf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf1c5252965e731d77af503529ca42e269c305c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf1c5252965e731d77af503529ca42e269c305c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
