)]}'
{"id":"openstack%2Fcinder~785668","triplet_id":"openstack%2Fcinder~master~I27d76a41297befc51f329fdca6f63bb19f253424","project":"openstack/cinder","branch":"master","topic":"add-placeholder-migrations","hashtags":[],"change_id":"I27d76a41297befc51f329fdca6f63bb19f253424","subject":"Reserve migrations for DB backports","status":"MERGED","created":"2021-04-09 15:35:18.000000000","updated":"2021-04-16 18:48:56.000000000","submitted":"2021-04-16 18:48:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785668-add-placeholder-migrations","meta_rev_id":"1a0022be74e9ee9eb098755f4fe1babf50d88897","_number":785668,"virtual_id_number":785668,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 18:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2021-04-16 14:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-16 14:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-16 14:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 16:34:59.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-09 16:35:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 16:55:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-09 17:03:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 18:50:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 19:39:06.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-04-09 21:02:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-09 21:47:38.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 21:58:14.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 22:19:21.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-09 23:18:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-09 23:47:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-10 00:10:07.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-10 00:15:57.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-10 00:44:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-10 02:10:12.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-10 02:47:41.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-10 21:56:02.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-11 19:20:20.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-12 05:55:22.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-12 08:16:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-16 14:21:40.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-04-16 14:24:31.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"d3ed19d5779f900b0f6af4cd8cb3ddc4dc83a156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-09 15:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd96b0a0685fd04d64a526bb0cbd7d20a37a5b5a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-09 16:34:59.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/785668/1/iscsi/ : SUCCESS in 58m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/785668/1/smb/ : FAILURE in 13m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1e8e4b7a2eb5775e7fc47d656effe2cc42ba68dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-09 16:35:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"8d084e955603fe8d53e1f4d5ea13b3a13cdbe238","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-09 16:55:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/68/785668/1/check/dsvm-tempest-cisco-zonemanager-job/5a70db7/ : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2dceba87783efc39ab309f87da37b7930c95d76d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 17:03:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/68/785668/1/check/iscsi-driver-primera/8bbc963/8bbc963.html : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":1},{"id":"06e6553151fafbaac809a168045116ac46032584","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-09 17:26:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/785668/1/check/iscsi-driver-nimble/e26391a/e26391a.html : FAILURE in 1h 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"95dd61d7301072858dddb8fb959b7ddf52e93186","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-09 18:05:19.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6a4404b277eeb78f23fbd743f72a66c0117bd6be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 18:50:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c6584d6d83d4f54a03efe1bb5eead3d : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36228eebf55843958968c0d8024f2b27 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aade4500da34bf7a764cce5e885c9da : SUCCESS in 11m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618f4d66bebe4bb6a7255d043c4d1814 : SUCCESS in 13m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79baedff80b443fc9fb7da3938bdf4c8 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb69821d61647c88d5fdb821110f0cc : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/20d7a0851cba47c08d7d5c2e2da139a2 : SUCCESS in 54m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3979225e9a6c4395a5c1ab575216645c : SUCCESS in 1h 05m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/600376dc919244b38367d31eafee097d : SUCCESS in 12m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/248372838acb4d3c9d3b5faedc2c4e02 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a81565fc42a044c399a68995af925af4 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4d78958ab4140a3922aa84a95b5aebd : SUCCESS in 8m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7eeca4ddd954bf5899ec1adb39090f5 : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5b3f6dc13e4409ca9380826f4ef7f7b : SUCCESS in 8m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/75691165c034411fba1b108615de62d7 : SUCCESS in 1h 18m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e6b1b944cf7d4ff5b1f9794d60ede3f4 : FAILURE in 1h 25m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0bf683202731407eac6af47567da73b2 : SUCCESS in 1h 28m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d505bdb0d6a14c998b85f61096f4231b : SUCCESS in 1h 05m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f05196656184b8ca97579dbba84252f : SUCCESS in 1h 19m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/93ade2c12d5a4546aeb09252f732fa6c : SUCCESS in 1h 28m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc1aaa2d6faf4313a9dadbcc743fe88b : SUCCESS in 2h 07m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55342d26bb5d417b8c41072f3209631f : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5a6c16a496b2858dc062e6e8adfe04aa1a0a551d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-09 19:39:06.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerStore_iSCSI/fb56f29/DellEMC_PowerStore_iSCSI/765 : SUCCESS in 46m 20s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerStore_FC/5c7b754/DellEMC_PowerStore_FC/760 : SUCCESS in 48m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"c88c32a00920a7724839030deff01e759c4c50a9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-09 21:02:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_Unity_FC/bb7ca64/DellEMC_Unity_FC/1107 : SUCCESS in 1h 20m 40s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_Unity_iSCSI/a218027/DellEMC_Unity_iSCSI/1044 : SUCCESS in 1h 41m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"704d5460c27fc7f2266e12c3428eae7e0a1bb95c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-09 21:02:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"133d4edcb0a32ee64b483b9e8b1849c8fcafc8d6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-09 21:47:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-785668-1/ns5_nfs : SUCCESS in 4h 53m 13s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-785668-1/ns5_iscsi : SUCCESS in 4h 57m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5a64b97a630e5f3a10e6ad02733461186f8f8a93","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 21:58:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_XtremIO_FC/a2484a6/DellEMC_XtremIO_FC/1278 : SUCCESS in 2h 01m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_XtremIO_iSCSI/45332d9/DellEMC_XtremIO_iSCSI/1300 : SUCCESS in 1h 46m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"92da6477086a6d685421190307089f42386d32f3","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-09 22:19:21.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_SC_FC/5a59ebe/DellEMC_SC_FC/607 : SUCCESS in 1h 57m 46s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_SC_iSCSI/d22d107/DellEMC_SC_iSCSI/604 : SUCCESS in 2h 06m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"81a3f73aba1b70891035dec42f33add124de05ed","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-09 23:18:03.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/785668-1/Seagate-iSCSI : SUCCESS in 44:22.30\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/785668-1/Seagate-FC : SUCCESS in 38:10.67","accounts_in_message":[],"_revision_number":1},{"id":"d06d5d2e68620111791a1eb1cb4dab5e98258015","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-09 23:47:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/785668/1/20210410062021/NEC-ISCSI : SUCCESS in 2h 11m 45s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/785668/1/20210410062021/NEC-FC : SUCCESS in 2h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c0e1351251043a95c5ac4adb91d62a610d674a93","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-10 00:10:07.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/785668-1/Lenovo-iSCSI : SUCCESS in 47:47.10\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/785668-1/Lenovo-FC : SUCCESS in 44:43.80","accounts_in_message":[],"_revision_number":1},{"id":"fffac60f34561f8871cc53a3ddf15e0f982bc2c1","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-10 00:15:57.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/785668-1/HPMSA-iSCSI : FAILURE in 1:03:46\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/785668-1/HPMSA-FC : SUCCESS in 33:36.05","accounts_in_message":[],"_revision_number":1},{"id":"8897a0d8caeaa39122554069d33c9d212ccab52d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-10 00:44:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerFlex/b9fbf2d/DellEMC_PowerFlex/753 : SUCCESS in 2h 40m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a7f6b74871bda834c1e2458f520306ef77290d9b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-10 00:45:52.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"1483b150fd921e12c643b911ba330dee02d5b6c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-10 02:01:59.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e67209ed32687b0c5474245eade904c5fcc8ec0a","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-10 02:10:12.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/785668-1/PVME-iSCSI : SUCCESS in 58:32.99\n* PVME-FC http://os-logs.tristero.net/dell-ci/785668-1/PVME-FC : SUCCESS in 56:31.06","accounts_in_message":[],"_revision_number":1},{"id":"154934b8edd12a4a7eb129c8dcf48cd478b6ba3f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-10 02:47:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_VNX_iSCSI/15f4aa1/DellEMC_VNX_iSCSI/733 : SUCCESS in 2h 52m 55s\n- DellEMC_VNX_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_VNX_FC/a22e2eb/DellEMC_VNX_FC/782 : SUCCESS in 1h 31m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"284593879adde5340c50a5f2a9f13e022be1f931","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-10 04:06:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d4a732c163d51d9b6df5d9d033ef298d44b56148","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-10 21:56:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/785668/1/check/tempest-dsvm-full-focal-py3/c7182c2/ : TIMED_OUT in 3h 09m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/785668/1/check/devstack-plugin-ceph-dsvm-focal-py3/91c6362/ : TIMED_OUT in 2h 37m 13s","accounts_in_message":[],"_revision_number":1},{"id":"29e2bb33fc705a236369caadf077df984931bb86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-11 00:22:58.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b674c75ede5d77ab46c1a37719e2d9171160df6f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-11 19:20:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerMAX_FC/299eb12/DellEMC_PowerMAX_FC/699 : SUCCESS in 6h 34m 18s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerMAX_iSCSI/db11520/DellEMC_PowerMAX_iSCSI/651 : SUCCESS in 5h 13m 07s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"c7e53711b2bb05ebad40c123e66085c0fd4e741b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-11 20:42:37.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b40ea0e8baa931adedbd3bb2d4749295e3611824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 04:16:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4cf5d4672aa4068ba61b1ecb8268fd8 : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b111d55c099e57ed0a8a2f328ec897a54801ad9a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-12 05:37:56.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerStore_iSCSI/d48b294/DellEMC_PowerStore_iSCSI/770 : SUCCESS in 52m 56s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerStore_FC/9fefb00/DellEMC_PowerStore_FC/765 : SUCCESS in 45m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"d39b1644140731ae3ad0edabbcd7217808041ecf","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-12 05:55:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/68/785668/1/upstream-solidfire-check/cinder-solidfire-iSCSI/d5f4caf/ : FAILURE in 1h 37m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"780480df5705dee511dc2d90617527b0b5b81c98","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-12 07:33:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_Unity_FC/b51ae6f/DellEMC_Unity_FC/1113 : SUCCESS in 1h 19m 14s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_Unity_iSCSI/ee5d16c/DellEMC_Unity_iSCSI/1050 : SUCCESS in 1h 46m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b9ce92f69972941ae27cdb3f4d8bd896aa02c9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-12 07:34:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"5c625c9971c67d17e5fa7938d2635f84fda79e40","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 08:10:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_XtremIO_FC/5a8acd0/DellEMC_XtremIO_FC/1284 : FAILURE in 1h 59m 06s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_XtremIO_iSCSI/22a3ee4/DellEMC_XtremIO_iSCSI/1306 : SUCCESS in 1h 44m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e02c8ff5fc7c609e8341c0ff958da5c11a5634c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-12 08:11:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"175a75b4df767c8b9587fabd7657b0bf31d39848","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-12 08:16:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/68/785668/1/upstream-check/cinder-cDOT-iSCSI/bbc371d/ : SUCCESS in 34m 36s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/68/785668/1/upstream-check/cinder-cDOT-NFS/6d5e51e/ : FAILURE in 2h 05m 18s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/68/785668/1/upstream-check/cinder-cDOT-FCP/fec5f04/ : FAILURE in 58m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ec202b50953a248073216a9c2ef46e0757dc64ba","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-12 08:48:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_SC_FC/baba2fa/DellEMC_SC_FC/612 : SUCCESS in 1h 50m 38s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_SC_iSCSI/84eb869/DellEMC_SC_iSCSI/609 : SUCCESS in 2h 21m 33s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"5644c8463c58da010dd70be923aef36dfc2cc448","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-12 10:13:37.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"64463421918e2a8c326086ca7224df6f97cbd7bf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 10:29:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerFlex/5bd94e2/DellEMC_PowerFlex/758 : FAILURE in 2h 54m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"91926f41965030530142baeb46a8b9be47543f6a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 11:56:00.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_VNX_iSCSI/269d2b9/DellEMC_VNX_iSCSI/738 : SUCCESS in 2h 51m 59s\n- DellEMC_VNX_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_VNX_FC/9c2305a/DellEMC_VNX_FC/787 : SUCCESS in 1h 26m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d43edb30dde514a36d34d0495bdf5d76c5ecaa26","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-12 11:57:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"b666e9ad1da42a3344257d3fc56fc04807660d54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-12 14:54:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"38ae59371c47b845ba18da47d69fd5b1398fdddb","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-12 23:09:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerMAX_FC/e5477ab/DellEMC_PowerMAX_FC/703 : SUCCESS in 5h 16m 48s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/68/785668/1/check/DellEMC_PowerMAX_iSCSI/de9a67c/DellEMC_PowerMAX_iSCSI/655 : FAILURE in 9h 53m 19s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"ed687d3f1b803590b3b1cd8e0a076c934373d097","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-13 02:12:59.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/785668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e2ac8a79cc941de6a91522540b68e7482b22e160","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-16 14:21:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0558a2b809852ffe29b2bca9e95a2b06eef3c9ed","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-16 14:24:31.000000000","message":"Patch Set 1: Code-Review+2\n\nI remember we used to reserve 5 migrations but never backported that many migrations so 3 seems like a good number to reserve.","accounts_in_message":[],"_revision_number":1},{"id":"f376c08422fa6f5b8d6dd6a337c14e801c864e4c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-16 14:26:49.000000000","message":"Patch Set 1: Workflow+1\n\nEric +2ed before my review so W+1.","accounts_in_message":[],"_revision_number":1},{"id":"6942d7ab35c486eca29144ac15cd11c5cbe06b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 14:27:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1147699b58d91558306f23ce5d2e7250ef8f8c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 18:48:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db7b4dc925d849259d9792041aa1e05c : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4a0753d12347939a513f038a94385a : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d9da437fc748cc96f4907706b7b467 : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36bc2d149e684c378570d039dd09717f : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f003f226f70441a9b711fcb744cf7ab8 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/527ed5e4b87d4158894082c063ce2d5e : SUCCESS in 1h 06m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/230bcc3f810f4cc48f4deca841528e36 : SUCCESS in 1h 11m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ed1057340544fc78d470eb066b7492a : SUCCESS in 1h 22m 46s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cdfdbbb8c352473399dd62eef02df523 : SUCCESS in 1h 20m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64ecb914781b4ec7bd30ffb41aa9fe07 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1a0022be74e9ee9eb098755f4fe1babf50d88897","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 18:48:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3475957b8a83ead762b18e296862f0efd731729b","revisions":{"3475957b8a83ead762b18e296862f0efd731729b":{"kind":"REWORK","_number":1,"created":"2021-04-09 15:35:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/68/785668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/785668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/785668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/785668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/785668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/785668/1"}}},"commit":{"parents":[{"commit":"e31158cd242468f11421b524c32a6bbdfc6136d8","subject":"Merge \"Add warning message about slow volume backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e31158cd242468f11421b524c32a6bbdfc6136d8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-04-09 15:32:04.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-04-09 15:32:04.000000000","tz":-240},"subject":"Reserve migrations for DB backports","message":"Reserve migrations for DB backports\n\nThis reserves three versions for potential backport to stable/wallaby.\n\nChange-Id: I27d76a41297befc51f329fdca6f63bb19f253424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3475957b8a83ead762b18e296862f0efd731729b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3475957b8a83ead762b18e296862f0efd731729b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
