)]}'
{"id":"openstack%2Fcinder~785970","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I482c3d5520a7bbb9971942cdb4c4208052106d70","project":"openstack/cinder","branch":"stable/ussuri","hashtags":[],"change_id":"I482c3d5520a7bbb9971942cdb4c4208052106d70","subject":"Backup manager: Synchronously call remove_export","status":"MERGED","created":"2021-04-12 20:29:18.000000000","updated":"2021-05-05 14:17:06.000000000","submitted":"2021-05-04 14:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785970","meta_rev_id":"b6810fb01c753c1467d632e1c8a0610f0fdeacbc","_number":785970,"virtual_id_number":785970,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 14:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-05-03 16:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-04 12:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-04 12:53:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-05-03 16:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-12 21:00:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-12 21:35:25.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-12 21:35:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-12 22:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 23:01:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-12 23:13:31.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 02:35:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 11:55:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 12:56:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-29 19:53:38.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2021-04-29 21:03:47.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 21:55:35.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 21:56:03.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-03 16:01:42.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-05-04 12:53:13.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-05-05 14:17:06.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"b0260d11865d847e91b58843ee03442e9b05a3ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-12 20:29:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91e4c99f20dbfc62d030575f14c115a9d6423fa6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-12 21:00:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/70/785970/1/check/dsvm-tempest-cisco-zonemanager-job/d78537a/ : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9d57305f0b02ff737b1ed6393ff97de358389d32","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-12 21:35:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/785970/1/iscsi/ : SUCCESS in 1h 05m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/785970/1/smb/ : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f43948e85e0f529ef956257fc82c8b2b6048754c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-12 21:35:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/70/785970/1/check/iscsi-driver-primera/3c22367/3c22367.html : FAILURE in 1h 05m 01s","accounts_in_message":[],"_revision_number":1},{"id":"234571afb3cd4af9a7feba9324d86d50ea3ee2bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 22:35:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d4fb7b3922647488c0739d7b18fb327 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8cd67f94c245de88ecf056ace7a244 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8a970685eb848de9fee38dac3cabd7f : SUCCESS in 10m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9915cb5f61dc435eb37c95f16b82d8ae : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4891875a17476081908d8e31820b33 : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/fc5a925a68664392b7957232c42df9e4 : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6472c385f74f4b8797ad79c2306f4276 : SUCCESS in 1h 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c6ccd286424aa2b37655d770f89a4d : SUCCESS in 6m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b2f7e15ad03f43f3b358313b0e6a4977 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27fce93b35aa49bb820af07035fb9d80 : SUCCESS in 6m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4a897bd23b9a43c3810daba69b015099 : SUCCESS in 11m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c782b15b2ba84709b420df86249738e2 : SUCCESS in 34m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47dc11764bf645409ac33f200b341a83 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efd2f74b79954717abeefcb9a7823e63 : SUCCESS in 1h 13m 16s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61b8129feac1467f9d4b607a0ce32ce6 : FAILURE in 2h 05m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/87967b90aa774289ae7d8968e03977bb : SUCCESS in 1h 06m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92379faf4f5c4f669756124ee3616d84 : FAILURE in 1h 44m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/532c8cf5af494c4dbfea51f3d72c5bc7 : FAILURE in 58m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11b89e28e031452183d85bb1ff9e8ab1 : SUCCESS in 1h 56m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db8ab941f1cb4759ba8735a58c5ab423 : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":1},{"id":"96f2e88320f91cc91e89e529017747755bd6f6f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-12 23:01:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a3cc4c5f9c2395ce5cd491d8b8f7b6b226713d75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-12 23:13:31.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/785970/1/check/tempest-dsvm-full-bionic/8959b3b/ : TIMED_OUT in 2h 38m 42s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/785970/1/check/tempest-dsvm-full-bionic-py3/c85c8d9/ : SUCCESS in 1h 50m 13s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/785970/1/check/devstack-plugin-ceph-dsvm-bionic-py3/38a9c46/ : SUCCESS in 2h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"65841dc477f3d4ce34ebec3dd31b21ff83cc23eb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-13 02:35:56.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-70-785970-1/ns5_nfs : SUCCESS in 4h 50m 56s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-70-785970-1/ns5_iscsi : SUCCESS in 4h 54m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"eda364344705fc444b6636ea501f307c1515fa7f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-13 02:36:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"ad2054f408cf0b5ad7639a733752c31438fb9b04","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-13 05:04:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"20d1930b7937182ff8ba97755a6d54fddcc32df5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 06:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/70/785970/1/check/iscsi-driver-nimble/a685445/a685445.html : SUCCESS in 1h 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e663a7255dc7ea2013cf192cf72d72bd5e55558a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-13 11:55:09.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/785970/1/20210413193517/NEC-ISCSI : FAILURE in 1h 19m 03s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/785970/1/20210413193517/NEC-FC : FAILURE in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6f3bd4666316944f3b920da8ce476c3ba223ddf7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-13 12:49:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"267f4b17387bf905854f9ee6c27eed61b75ad995","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-14 12:47:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c49541b3763af03bff4cc408e25e6c7f342cc60d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-14 12:56:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/785970/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6290009/ : FAILURE in 6m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/785970/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9db7ce4/ : FAILURE in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8cdf0ed0b8f2cb1d5837726853a6a74ee55b2487","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-14 13:20:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/70/785970/1/check/dsvm-tempest-cisco-zonemanager-job/b54a0c2/ : FAILURE in 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"00c1739972d0e1f4c400f8763cc8ed34d3f427c9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-14 13:58:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/785970/1/iscsi/ : FAILURE in 1h 01m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/785970/1/smb/ : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":1},{"id":"674dd329a0e66be4398700dc15f4ffeb1f55306e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 13:59:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a731b7857aac865654caa36040438368106e5827","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 14:49:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2ebe7dc0581078ab89321d5405d95025bc8019e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:55:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ebe16c23ba84cdaa4b90fc5c0c9099b : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/335f635106b54a3c998c0d530cbb4f94 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8477ab3af9ac4b178240e96154136e50 : SUCCESS in 11m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c7a5420ed594775827c775ce1a25d83 : FAILURE in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c96e67b065148329d41a7204168c2e5 : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/c4cb8f4d82b240698ceb90db0ba1966e : SUCCESS in 51m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9147638a8ba54056a788d06e32c2ebd8 : SUCCESS in 59m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc3263a17e0d4fdd8b53a4d798fae2b7 : SUCCESS in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae451e2043594d108d9426e1fbd4d4ee : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4bb8b0c024d49609bdd6c7943a65804 : SUCCESS in 9m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5840dad439a04b1f8078847a4aae39a7 : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/567c8b2fc2e241be96982dd31e0a85f0 : SUCCESS in 47m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5e5d1da35948453b9218e32c798a5733 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/00619222b426449ba1d10dcf71c90ef8 : FAILURE in 1h 18m 18s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ccc50f3ae39e4da5a94cfcac8634c645 : SUCCESS in 1h 36m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0155f485a8c94014bd9e659e2962605b : SUCCESS in 1h 30m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f249daef0cf640778e2db37c2e28a944 : SUCCESS in 1h 38m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc87edeb1c604028b80d66c6f08debc0 : FAILURE in 1h 11m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a2e2225d6de4dd99998e74f1d9f5f28 : SUCCESS in 2h 06m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2d82caefa5f4e539d7b24e75e3c3d89 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d62b2f0b51d46233d613374781be62273f68d20e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-14 16:35:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/785970/1/check/tempest-dsvm-full-bionic/262867b/ : SUCCESS in 1h 56m 31s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/785970/1/check/tempest-dsvm-full-bionic-py3/392fac0/ : SUCCESS in 2h 07m 20s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/785970/1/check/devstack-plugin-ceph-dsvm-bionic-py3/f444a72/ : SUCCESS in 2h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"874291efc95bf4ea3f2babfdfe1e3a7474c11ff3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 18:40:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/70/785970/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1deb13fb0e0592f5ebacfba7c580fd9aaa941a18","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-29 19:53:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"296271105a31b4d5a2fb75dc4cc5ff7efdc5231a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-29 20:01:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/785970/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5640b2a/ : FAILURE in 6m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/785970/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ad67fa4/ : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ae62f48a174b89cbf87dc682394849e84b420d89","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-29 20:27:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/70/785970/1/check/dsvm-tempest-cisco-zonemanager-job/9608106/ : FAILURE in 28m 21s","accounts_in_message":[],"_revision_number":1},{"id":"21288719f81ee1f069f26193b4ba1efc636ed38f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-04-29 21:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/70/785970/1/check/dsvm-tempest-zadara-driver/b5a9e31 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 09m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6b1ead8526eed4c97edc81854752be659f6b860","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-29 21:12:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/785970/1/iscsi/ : SUCCESS in 1h 17m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/785970/1/smb/ : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"36bc65e1d12b89c6b39b87ccccd9f4f48a981881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 21:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c4a5994b774bac920b896cf13e285f : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f614ab6ef84cf9adc5e1e7b648169d : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc0399acc1124363b2c8430597b68001 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4b7e57e5004d74a351786879f25920 : SUCCESS in 14m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b8a77f32cd7743b8be4d434d21084521 : SUCCESS in 1h 05m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/901980d98d2044fc98871886599374d1 : SUCCESS in 1h 01m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b00039e9528458c8f2fb435a0be26b5 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de9f8993a2b84ae28edd648010ff365c : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28544ed7c07449338a24e9794aa02bfd : SUCCESS in 5m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7e9e103d50fa4cc69e7103337c1a0a8c : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f4ef97576ad44b98af41ad24c55e4d2 : SUCCESS in 41m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b2788c7e7bd4f18810a92e93d179383 : SUCCESS in 6m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fe4d224eade74371ad7c4366060a7506 : SUCCESS in 1h 13m 36s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6be508c056644bd1b84eb41d767e6a46 : SUCCESS in 1h 18m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a17ec42e411448dd99f06bc2a4eebd1a : SUCCESS in 1h 26m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f58a6a69d5a46f4945b69eaca78f8eb : SUCCESS in 1h 21m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/59632bd24b524545b50c67209bc5e821 : FAILURE in 1h 07m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/606b31f30fb64089a2125dc2ed71796a : SUCCESS in 1h 56m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ff3f4bbb7084db7954a07966eb9b7a4 : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":1},{"id":"74f4078b1ec571b68c428f4b732799ab32549ac7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-29 21:55:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/70/785970/1/upstream-check/cinder-cDOT-iSCSI/fb51328/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/70/785970/1/upstream-check/cinder-cDOT-NFS/f668f4c/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/70/785970/1/upstream-check/cinder-cDOT-FCP/f038440/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b56ef5ed130af623113f01a89018865f04caed44","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-29 21:56:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/70/785970/1/upstream-solidfire-check/cinder-solidfire-iSCSI/abfdfa2/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"902ffc7d6157faab1e5a835b64460ff39d20d8ce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-29 22:41:13.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/89681e0ec8dc48a58177096ced5d84b0 : POST_FAILURE in 2h 43m 32s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/afa445347384468994baa000bf1a587a : POST_FAILURE in 1h 44m 10s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/0666baaf82cc4578b27dbecfb1332500 : POST_FAILURE in 1h 58m 41s","accounts_in_message":[],"_revision_number":1},{"id":"cd2c451d9b6016f367a12400a6de5fa16ef4f23a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-03 16:01:42.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\nThis should be backported.","accounts_in_message":[],"_revision_number":1},{"id":"54a55cae676252d876e43926c64e1884b59941fe","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-05-04 12:53:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks like a good backport.","accounts_in_message":[],"_revision_number":1},{"id":"4529ec838e92338bcd0211a38bb5d149ba9f9053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 12:54:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"110d789dcf0914bbb75e8b77b5619472b1fc56ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 14:24:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260afd763f4a45479b00aa31b89e4668 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/282952279ac44c85b92dc0555414e1eb : SUCCESS in 9m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19235ab48dd143e88b10c4781b43af18 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151a5692732a47e188c8392d2ba1fefe : SUCCESS in 15m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/9b09fe74179c45b6be596b687514c298 : SUCCESS in 1h 16m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/041a4866cb7e451fbe2863a74e394f66 : SUCCESS in 1h 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a039f455768347c1a57c229ffe2943af : SUCCESS in 6m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef2493717f66497e9e7d4df1b75c4e43 : SUCCESS in 1h 24m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6bcc5afa4ef7490c8a25edf8de0b4567 : SUCCESS in 1h 17m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da53b84c2173422a96442d2157124017 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b31a13591687e0da01978967f2d298c69ec76e61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 14:24:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aceab9ddbc44963297cfe356d74f4aeb2e83ba8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 14:26:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e6b1aabe904a8cbd8aec356432607e : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a09b3c8e78a44c49ac64a6e9af6fe954 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b6810fb01c753c1467d632e1c8a0610f0fdeacbc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-05-05 14:17:06.000000000","message":"Patch Set 1:\n\nShould this backported to train as well?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34120d49808537d0523375b173db9c657114a4cb","revisions":{"34120d49808537d0523375b173db9c657114a4cb":{"kind":"REWORK","_number":1,"created":"2021-04-12 20:29:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/70/785970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/785970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/785970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/785970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/785970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/785970/1"}}},"commit":{"parents":[{"commit":"1aa9073008083c35fa5ac6e44cfcafbce27a60db","subject":"Pure: Add default value to pure_host_personality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aa9073008083c35fa5ac6e44cfcafbce27a60db"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 16:01:33.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-12 20:29:10.000000000","tz":-240},"subject":"Backup manager: Synchronously call remove_export","message":"Backup manager: Synchronously call remove_export\n\nCall the volume\u0027s rpcapi remove_export method with\na \"call\" instead of a \"cast\" when detaching volumes\nin the backup manager.\n\nThis prevents _detach_volume() from returning before\nthe remove_export() has completed, which helps prevent\nproblems where remove_export() races against a\nsubsequent operation.\n\nCloses-Bug: #1920237\n\nChange-Id: I482c3d5520a7bbb9971942cdb4c4208052106d70\n(cherry picked from commit d6895f9d2aab84c116efc2f853bb5f9e57fac539)\n(cherry picked from commit 5402ac8e2d30872b526c3ac7d9f10645452fbf66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34120d49808537d0523375b173db9c657114a4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34120d49808537d0523375b173db9c657114a4cb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
