)]}'
{"id":"openstack%2Fcinder~786054","triplet_id":"openstack%2Fcinder~master~Id1f111462fc2558b90f9066387498bcb5e3217b2","project":"openstack/cinder","branch":"master","topic":"new","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-03-02 11:54:20.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-25 13:30:32.000000000","reason":"Brian Rosmaita replied on the change"},"16186":{"account":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"last_update":"2022-03-02 11:54:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1f111462fc2558b90f9066387498bcb5e3217b2","subject":"Move nimble driver code to hpe folder","status":"MERGED","created":"2021-04-13 13:53:34.000000000","updated":"2022-03-02 12:26:40.000000000","submitted":"2022-03-02 11:54:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"786054-new","meta_rev_id":"56380d2c383bc5a6e87c68cfe55672f1e06fecc0","_number":786054,"virtual_id_number":786054,"owner":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 11:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-25 13:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-10 13:13:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":2,"date":"2022-03-02 10:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":1,"date":"2022-03-02 10:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-25 13:30:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 14:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 14:48:47.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-13 15:52:11.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 16:23:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 16:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 16:44:23.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-13 21:08:24.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 21:31:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 06:58:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 08:26:31.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-18 20:16:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-04-19 06:28:13.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-19 07:06:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-19 08:34:03.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-19 17:08:32.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 03:10:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-04-20 07:32:21.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-04-20 09:20:37.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-20 09:39:05.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-20 10:01:59.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-20 10:18:13.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-20 14:13:28.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-20 18:22:29.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 22:07:05.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-23 07:15:06.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2021-11-23 18:58:57.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-24 12:31:50.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 10:19:43.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-01 14:45:32.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-02 14:02:45.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-12-02 14:29:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 14:03:09.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"REVIEWER"},{"updated":"2022-03-02 10:31:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"e467cb360e809892df459396f5ada4e3530459f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-13 13:53:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"543e31e184615763b2754fc36689830f77b4b3c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 14:26:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ee52f0350a14a3d93fcd48cd991a1d9 : FAILURE in 14m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fc0e7d8f85e446299f65f66f2df5227 : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea01af2200e692bef2535891fd065afdd952097f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-13 14:48:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/1/iscsi/ : SUCCESS in 54m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/1/smb/ : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fd5aa7dcb85686adceaecd911ae533947343c752","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-13 15:52:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/54/786054/1/check/DellEMC_XtremIO_FC/085cdab/DellEMC_XtremIO_FC/1297 : FAILURE in 1h 58m 06s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/54/786054/1/check/DellEMC_XtremIO_iSCSI/6103a14/DellEMC_XtremIO_iSCSI/1319 : SUCCESS in 1h 45m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"c69f9676b9b03580831b827a2984c31553c211e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 16:23:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/54/786054/1/check/iscsi-driver-primera/04d9c51/04d9c51.html : FAILURE in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bea9ec8cb6dc2a86a197074ab6d95f79c7562ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 16:40:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7dc839d8e9243a9871f381208ca8d52 : FAILURE in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c917796c4bbc4f74b5bb0a6efa5e9997 : FAILURE in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3257ce02457e4122acd5f8f6045023e5 : FAILURE in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5d5388eb2054a93a9d6e1c8c58f7a11 : FAILURE in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8388e71372e54d848aa34598a0564e98 : FAILURE in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cfb95c8354f43f7accc611d56ebdb14 : FAILURE in 6m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/a392b188ea0647cca8972f7009ba0dc1 : SUCCESS in 59m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/418ec128e460414789268a6829382f52 : SUCCESS in 55m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6f86d1c017b41958bbae9a808f1f6e7 : SUCCESS in 6m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb50cb9f5c8144f5a2e7ec797cff41ad : FAILURE in 5m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b9bbe6bd09b4cb8a64c0e89f4bec4ba : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92cbd716ed8146019320c75446a4b112 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07ce84ae62ec456591fecdb5a79ffaf6 : SUCCESS in 13m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f0ba8dad81114656ae4ff629cc226f40 : SUCCESS in 30m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/582b2bcdb5ac47319dcc7be59e5f67d4 : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/34981d1c0ff140cc87b55dc65be968cc : SUCCESS in 1h 11m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b50150e647843d0ab732075ebe98492 : FAILURE in 25m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7b23045733bc4fb093111d6857e186a6 : SUCCESS in 1h 20m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e78385d4916f4cd2a580ffb1c1aaa4a8 : SUCCESS in 1h 26m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbd42fb9ace947e8bbc49f469c5728d7 : SUCCESS in 1h 04m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8de9f030993d40188ffe877f603a5d71 : SUCCESS in 1h 15m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a58f270d81054dc09dd14e96c8be5037 : SUCCESS in 2h 05m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc7314f8defc4c96bc8906f627ece83e : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":1},{"id":"470575533b0165ef9894f237a60714d0ea5e4960","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-13 16:44:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/1/check/dsvm-tempest-cisco-zonemanager-job/49f4316/ : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7c68a9870e89203d5e260217c883fa0808a76250","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-13 21:08:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-1/ns5_nfs : SUCCESS in 4h 49m 56s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-1/ns5_iscsi : SUCCESS in 4h 48m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c10ea529446ad114b6adc8e2850b309f9ed47ec2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-13 21:31:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/1/check/tempest-dsvm-full-focal-py3/4b34e91/ : SUCCESS in 2h 04m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/1/check/devstack-plugin-ceph-dsvm-focal-py3/727ee11/ : TIMED_OUT in 2h 43m 49s","accounts_in_message":[],"_revision_number":1},{"id":"66118080aae98d8f98ec4051d5e750d244c1ba6f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 22:12:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/1/check/iscsi-driver-nimble/224f7e1/224f7e1.html : FAILURE in 1h 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3327f0ce807f4b930fd427e3bef80539cf03c4e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-14 06:20:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e71765910abbeb13c49833edf536c5c9af896ba9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 06:58:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/532e2bef68f44f7e81fe56378bda6be7 : SUCCESS in 29m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09b85c9b7d2d4cebb682dbdca83977d0 : FAILURE in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42e2319a95777e8018267f7aac9d2b0a345466a8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-14 06:58:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/1/20210414132128/NEC-ISCSI : SUCCESS in 2h 08m 42s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/1/20210414132128/NEC-FC : SUCCESS in 2h 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"32e195d89ea996f951226d6ff14efe625d4c441d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-14 07:18:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/2/iscsi/ : SUCCESS in 57m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/2/smb/ : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":2},{"id":"38893396cbfd621a95b79684c8dfe29a6a405d07","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-14 07:39:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/2/check/dsvm-tempest-cisco-zonemanager-job/89906a0/ : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":2},{"id":"3102e27b99f1e750c9ed4f5f8e4064f82db367b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 07:43:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/2/check/iscsi-driver-nimble/c94a49d/c94a49d.html : FAILURE in 59m 28s","accounts_in_message":[],"_revision_number":2},{"id":"50f592ad6a0231c2e65c2e710011001069a4f553","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 07:57:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/54/786054/2/check/iscsi-driver-primera/8ae6872/8ae6872.html : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fb873ab3ab1d2a8e3041ab6e48aa443dfdf6036c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-14 08:26:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/74dd59a/ : SUCCESS in 1h 53m 47s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8dff313/ : SUCCESS in 2h 04m 12s","accounts_in_message":[],"_revision_number":2},{"id":"80bd233d44103828a759fe619af93d6716ab9d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:58:43.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55857b88a8bc4afea9f7ef4b09c37e2a : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afbe9b5ad1b842caaaaf98bd63abf4c3 : SUCCESS in 17m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/729e0ff09c894beeb8704456b39deba9 : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33efa137ae2041a89e5e38dad0253b08 : SUCCESS in 12m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc3359b81eb94b5b993fca35193695d2 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a09de78b3449acb82901262f88bfe2 : FAILURE in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6babefcf44cf42bb9e12223e15f74e95 : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84b4e895127b4e389f9dd7d38bd38e6c : SUCCESS in 1h 07m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8290874f0d8e4392b19f4b66319acc80 : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6383717452d7437ca620a12f10af5a04 : SUCCESS in 14m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11301b3c39d74cf7acdb8e1f54b95c6a : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ea09716714c4dfcbc8849fe70dff4c4 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/293e848ab89f4a55ad6937c81b3cdc69 : SUCCESS in 7m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d33d1e136d504bd9b96f4ba76a746999 : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e3797c0f3c3414bb5cf76e3c3fca2b8 : SUCCESS in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6d99492bd774365bed6f8cad3d69d93 : SUCCESS in 1h 18m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5e665849ada47faa80021bda0ef40e6 : FAILURE in 1h 25m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f20b2f62735487faa779679fbc754c2 : SUCCESS in 1h 23m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed5cbfc16569449d93278698f455b21e : SUCCESS in 1h 09m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a55a5fdcacae418785be30267c087a59 : SUCCESS in 1h 32m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fea2a9ab02ae45e08bb46319502fbeb0 : SUCCESS in 1h 21m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce721813953d40d6a39b722b96ac1c09 : SUCCESS in 2h 02m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6360c7a187a54e8cb88f822f1520553a : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4116dc6c623f84be15442b5bd7ef8a16cdfdd9c9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-14 09:23:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/54/786054/2/check/DellEMC_XtremIO_FC/ad175a0/DellEMC_XtremIO_FC/1301 : SUCCESS in 1h 58m 34s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/54/786054/2/check/DellEMC_XtremIO_iSCSI/7b8e30b/DellEMC_XtremIO_iSCSI/1323 : SUCCESS in 1h 42m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7f9f60f153058f22be54fdbab54aee6aeccc75d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-14 11:33:59.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/2/check/tempest-dsvm-full-focal-py3/f0a9598/ : SUCCESS in 2h 00m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/2/check/devstack-plugin-ceph-dsvm-focal-py3/61a2538/ : TIMED_OUT in 3h 13m 08s","accounts_in_message":[],"_revision_number":2},{"id":"30aa7041d262536ba151e68a29a617c908c2277d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-14 11:55:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-2/ns5_nfs : SUCCESS in 4h 53m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-2/ns5_iscsi : SUCCESS in 4h 33m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"c3e6344261caf7681215b96f05e49041c2846ee3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-14 20:44:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/2/20210415032426/NEC-FC : SUCCESS in 2h 17m 57s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/2/20210415032426/NEC-ISCSI : SUCCESS in 2h 19m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2c24c62a773ce31dabdb7682a81d9d9e8aa6520c","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-17 03:25:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad7af864d9b04ec299e594a88992c5b559332511","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 03:48:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f61a8a2cf20d4c32b9fdd9633b61cea5 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76659c3027134e4886997bde0e9ae763 : SUCCESS in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb845f4092d18553a0b7ecfc0f03eacc34358251","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-17 04:20:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/2/iscsi/ : SUCCESS in 54m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/2/smb/ : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":2},{"id":"274560a4939327adb21cf00146e003001c956690","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-17 04:43:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/2/check/dsvm-tempest-cisco-zonemanager-job/5ef9b7a/ : SUCCESS in 1h 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a86ba9468cf751d806737282611daecb7e375347","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-17 05:01:36.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/58c74aff9d8a44788ca5ef309f62c388 : POST_FAILURE in 1h 32m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/041f48370caa4f3289a345e544337775 : POST_FAILURE in 1h 16m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9773a78034eeee4890a0dd5e5e7e89451be89218","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-17 05:07:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/25e44d5/ : SUCCESS in 1h 40m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0162b89/ : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b10137da465f6b91084eea04c786ce1e8d009d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 05:30:20.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93d07a8a9d66401c9404ee7a05c48efd : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f5659e308cc4b0d8bb5c4f79134350e : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3f43ac224674a4ea78b2a6bf6c04823 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2a2cc6793df4d1fb309ace630a81b42 : SUCCESS in 9m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e43d387fcb44eb19403b30fb69a3c1a : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe0c978ea2948a0bd008ebf96669c8a : FAILURE in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ef510ed301814055af78bc168f1febf1 : SUCCESS in 55m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e77beea2d6b9420dbc49dfa4ab0c2dad : SUCCESS in 1h 47m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc7d1fce963443379c7643ae4ad506f5 : SUCCESS in 6m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/46b055358e834d7cb3725577ddcc83e6 : SUCCESS in 12m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89c5b7cbdf4c4ca48c5f83674de2368c : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f0f8e38f65f40158fed010a1be4f4ec : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7e35d7a3cc894fba97adb5675db38342 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b7b69ad3f6464859af56ef1200722e6d : SUCCESS in 31m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65a969cd3a3b44df952a3be651797610 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94ca93ed85944b35bbb9ec0b8afbe656 : SUCCESS in 1h 12m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9cf5af11818430b862a84e6d0c3497e : FAILURE in 2h 03m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/259198cb1d5b4f22a4ef07f83c5ea407 : SUCCESS in 1h 21m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ddde23aaeda4e5c940fd58fdabd81af : SUCCESS in 1h 05m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac28c7eb88b84890bf0e08e9457c3529 : SUCCESS in 1h 05m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d1ea1ab1e64748b0b0a86bc1ca1ba84d : SUCCESS in 1h 54m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7807dabec2e5421db7e357c641a67c6c : SUCCESS in 1h 55m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e37bfee10cb74a3ebea773784e1941a6 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":2},{"id":"35f44a0be9f40d6a429a108e4f8cd0e80b38db4f","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-17 09:40:51.000000000","message":"Patch Set 2:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":2},{"id":"38ceddfbe756477365fae5767a69267e538a44ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-17 11:29:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/2/check/iscsi-driver-nimble/4bdb539/4bdb539.html : SUCCESS in 1h 47m 41s","accounts_in_message":[],"_revision_number":2},{"id":"48814081e8b3a001cc35033d2c74236361572538","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-18 20:16:39.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSee comment inline about the docs job.","accounts_in_message":[],"_revision_number":2},{"id":"e0c484ef7ca6f8f075980c17125ba95d88856508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-19 02:51:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ca229753b88ad6bcf1be9ae03f527290bf0f639","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 03:13:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7acf61adfbea444caf14693a8671f582 : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cb44d40cc2f491fb1f6ab272d6a173a : SUCCESS in 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1af319663bf32421f8db3b91cca9cd0ad9097368","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-19 04:12:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/3/check/dsvm-tempest-cisco-zonemanager-job/a8dddce/ : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":3},{"id":"c91806a49f10414575832ee5e297814f83acabc2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-19 04:16:53.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/3/iscsi/ : FAILURE in 1h 24m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/3/smb/ : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":3},{"id":"feff6e225f5bd4f4e2b5ff3cf5e3d872e04405d4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-19 04:29:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/654a980/ : SUCCESS in 1h 36m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/660c7bb/ : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":3},{"id":"dda8b7ca6577b8bdf57be04b0739b7a86851dd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 04:53:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a67a353145948abb79a91d2236d5c3c : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac0f894a098846bc95b41208c4b673ee : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/777802ddef504f4f90044f4582404915 : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5be9b691094745708c8c15f7aa974210 : SUCCESS in 12m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e35f78894d364af9b0ab245deb07dcdf : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9765c2bf16f4aa5b1fb944792974352 : SUCCESS in 16m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/a8d9666a717f40b0b37c4052358fa3d4 : SUCCESS in 1h 03m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be656ae4a71c435a887a5370ae513ce5 : SUCCESS in 1h 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdf992dd081a446fadc21e41b18fb812 : SUCCESS in 5m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84ca6d8b8d2a4872a94ba401f1e3cbe1 : SUCCESS in 16m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/18cc36c900c14607bd812329e37e4a99 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5486d2af760049bdbc7a16e385876cec : SUCCESS in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6787a1a1aac845f7ab9e63ec2c070a54 : SUCCESS in 9m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ea3ad770ae843e8a870fb83a6d34b61 : SUCCESS in 51m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a0de2321a7242a9aa63289797a7c773 : SUCCESS in 8m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97d3a110315c43fa916e2f0268ab8f2b : SUCCESS in 1h 47m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8dbca1aba37b4ae18eb8dc64be0a7772 : FAILURE in 1h 28m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fdf5a44f29244bdbeaba7f40ff0e60a : SUCCESS in 1h 57m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e76d1b334e445a1b4f63082cbfbc493 : SUCCESS in 1h 30m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/59878811d18d4be8be0adce382685267 : SUCCESS in 1h 30m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f613a03f49e34bb586c9946ec5f8e2b4 : SUCCESS in 1h 27m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2bc5e25ae8224d748b0e9590ad0f42a6 : SUCCESS in 1h 28m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a9346a0ae74284bb83833f8a4c0d5a : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9fa01c87453cc4e76a94d45b3648a877875b25a8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-19 06:18:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/3/20210419125712/NEC-ISCSI : SUCCESS in 2h 17m 16s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/3/20210419125712/NEC-FC : SUCCESS in 2h 20m 48s","accounts_in_message":[],"_revision_number":3},{"id":"1105bee016d2f3795030314889631a68a76fa904","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-19 06:28:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/54/786054/3/upstream-solidfire-check/cinder-solidfire-iSCSI/8cab5cd/ : FAILURE in 1h 33m 42s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"1f7c9a690ca0b4a0dc4767714500211bd3b8e1ab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-19 06:30:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/3/check/iscsi-driver-nimble/180d02d/180d02d.html : SUCCESS in 1h 52m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8bc8228c4c3fad11ed08db85f0aad0ab69f1c8dc","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-19 07:06:31.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786054-3/HPMSA-iSCSI : FAILURE in 1:02:24\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786054-3/HPMSA-FC : SUCCESS in 33:41.47","accounts_in_message":[],"_revision_number":3},{"id":"2800a4ee2b639520f0a4e5ef153ba17682cd23c0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-19 07:32:24.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/3/check/tempest-dsvm-full-focal-py3/d53f3e0/ : SUCCESS in 1h 42m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/3/check/devstack-plugin-ceph-dsvm-focal-py3/022fa9e/ : SUCCESS in 2h 07m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ff892c1214ee889255dce4c4a874aa4392b8034a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-19 07:32:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-3/ns5_nfs : SUCCESS in 4h 40m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-3/ns5_iscsi : SUCCESS in 4h 32m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"3b38c1816150cc0b9250784ebb63a07cd5c46299","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-19 08:34:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/54/786054/3/check/DellEMC_PowerFlex/ed772b6/DellEMC_PowerFlex/1 : ABORTED in 3h 40m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"e6b834bc81d3b5482119ff1ac866a364d81e7dd9","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-19 08:47:34.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e See comment inline about the docs job.\n\nThank you.","accounts_in_message":[],"_revision_number":3},{"id":"2dc3218613606881ca90144e45193db2b285a983","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-19 17:08:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/54/786054/3/upstream-check/cinder-cDOT-iSCSI/715e0ae/ : SUCCESS in 33m 55s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/54/786054/3/upstream-check/cinder-cDOT-NFS/1eb38b0/ : SUCCESS in 35m 48s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/54/786054/3/upstream-check/cinder-cDOT-FCP/986e0e6/ : SUCCESS in 54m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"638c82b0c6622fdc38aeba2ac2acfb8fc83c47c3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-20 03:10:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nSee comment inline.","accounts_in_message":[],"_revision_number":3},{"id":"5aadddbfbfc61baa681fb3aa28e73c90dabfb16d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-20 05:03:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1be389eff75110b788c13f127d9bf2dc3ae6ae60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 05:29:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a55709c1089e463a9a8a0d972eff0588 : SUCCESS in 22m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a4d8caaef4f4da09066d4ce0609eedc : SUCCESS in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a5dbea0e3d3257dc559102ecec717b131a9d9735","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-20 06:18:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/4/check/dsvm-tempest-cisco-zonemanager-job/b17825d/ : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":4},{"id":"b2b93b7dbcafaed23e5fe7e69ba0de7981a8c151","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-20 06:30:51.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/4/iscsi/ : FAILURE in 1h 26m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/4/smb/ : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4660165156f48461f2a6e54b6fddae63a69fcd2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-20 06:39:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f85430e/ : SUCCESS in 1h 25m 58s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5f45543/ : SUCCESS in 1h 33m 49s","accounts_in_message":[],"_revision_number":4},{"id":"f76190beda5d3e6ea8e3d944b402d54f84c93e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:15:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8839360412f475d8b3e523e85dfd00a : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4022c8671bd94191b21a5b22c54229a3 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe48184464884882abaeb7b2ab697757 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91c4f717a6f4472bbdda2d8236fe154e : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/609c3ae4035c4473b251a40ba1f7f7fe : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23e8e249e2f4b4b82823d153df28dad : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/4b49dc4991904598b6a71f0f9eec02cd : SUCCESS in 58m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f1f97e62d29464e915ad38a0c32ba97 : SUCCESS in 1h 05m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9596fb9c73b476eaf28c78b2d3bebf1 : SUCCESS in 7m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d2ad1c0e7bed4b51a756511ff54dbf12 : SUCCESS in 11m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da9b3ff8ba8348508d0a61c8b20f8166 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5aa093c5d1b847cdb00d01b10a6063aa : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc7ed5341ae54b4692432468a2450467 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/97168db8cd994dc1a5839907cce0fd94 : SUCCESS in 32m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70503a6bcc914fd4886b940c3e79b65f : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cac06b777aca4c72a69a1434198a9087 : SUCCESS in 1h 18m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eaa9af506db440e5833e6b2863ea4b75 : FAILURE in 1h 35m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71a3e8584cf148899badc10f32f99a31 : SUCCESS in 1h 42m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da971223ff3c466b9b3207d8b4e0b825 : SUCCESS in 1h 21m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b43b3bdbbdef4bc49bd6cda48b0ae14a : SUCCESS in 1h 18m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cde11148cbe542daad674891ac79d90d : SUCCESS in 1h 21m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40557cae78da4f9e88f2e4da3d8ebd48 : SUCCESS in 2h 05m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/069ece0a462046849dd4bdb7531f79b4 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":4},{"id":"91df3028abda294a096ea8d56c45caab50d5cd80","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-04-20 07:32:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-54-786054-4/1618902957 : SUCCESS in 16m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"9f8537d67125a39849ed22aafab7901d8582e2c1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-20 08:39:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/54/786054/4/upstream-solidfire-check/cinder-solidfire-iSCSI/9d0447c/ : FAILURE in 1h 22m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"dc0edb9f26486e5a7fb10c6afe96a02f3ff1b4eb","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-20 09:20:37.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786054-4/Seagate-iSCSI : SUCCESS in 47:43.62\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786054-4/Seagate-FC : SUCCESS in 39:37.08","accounts_in_message":[],"_revision_number":4},{"id":"9bfe4eb13f2b77ba08a0dc5adf28fa3ee618c32b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-20 09:39:05.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786054-4/Lenovo-iSCSI : SUCCESS in 49:53.96\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786054-4/Lenovo-FC : SUCCESS in 44:40.59","accounts_in_message":[],"_revision_number":4},{"id":"5477c74edc393792c472030e95074c3798b81525","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-20 09:53:51.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786054-4/HPMSA-iSCSI : FAILURE in 1:08:20\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786054-4/HPMSA-FC : SUCCESS in 33:26.75","accounts_in_message":[],"_revision_number":4},{"id":"bacb193a33c6fe7e88d5ddd571e89a8f2a8987c1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-20 10:01:59.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/54/786054/4/check/DellEMC_Unity_FC/2f91190/DellEMC_Unity_FC/4 : SUCCESS in 1h 25m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/54/786054/4/check/DellEMC_Unity_iSCSI/b9b77e5/DellEMC_Unity_iSCSI/4 : SUCCESS in 1h 32m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"341c9855d764372b79271bb58d7bb13d019b58b6","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-20 10:18:13.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786054-4/PVME-iSCSI : SUCCESS in 1:00:27\n* PVME-FC http://os-logs.tristero.net/dell-ci/786054-4/PVME-FC : SUCCESS in 52:55.29","accounts_in_message":[],"_revision_number":4},{"id":"e71772502946010b336728b63149be4638ad6f11","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 10:27:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4627af96a4357221f9af6c88b1682bfd69b0cae3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-20 12:26:46.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-4/ns5_nfs : SUCCESS in 4h 54m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-4/ns5_iscsi : SUCCESS in 4h 38m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"bae859cdb3847bb57cf55087672574398d3b5341","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-20 13:09:20.000000000","message":"Patch Set 4:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":4},{"id":"e9eadb9cc766f921279a3aad52be60ca8cf849a5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-20 14:13:28.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/54/786054/4/check/DellEMC_SC_FC/a97bd29/DellEMC_SC_FC/7 : SUCCESS in 2h 09m 01s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/54/786054/4/check/DellEMC_SC_iSCSI/4100626/DellEMC_SC_iSCSI/7 : SUCCESS in 2h 08m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"5cd84e6130d814312d762ea58722adb511ac179f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-20 15:12:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/54/786054/4/upstream-check/cinder-cDOT-iSCSI/b527a8e/ : SUCCESS in 33m 07s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/54/786054/4/upstream-check/cinder-cDOT-NFS/91391e8/ : SUCCESS in 1h 29m 48s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/54/786054/4/upstream-check/cinder-cDOT-FCP/c7fabd2/ : SUCCESS in 56m 09s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"04938a449459cf7ecb2b009076826ecbfd4836d0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 16:09:36.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/54/786054/4/check/DellEMC_PowerFlex/c5412d6/DellEMC_PowerFlex/7 : SUCCESS in 2h 32m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"e894c2100bf3605c604390ca9e913ef1c3228bc9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 18:21:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/4/check/iscsi-driver-nimble/dee0845/dee0845.html : FAILURE in 1h 50m 15s","accounts_in_message":[],"_revision_number":4},{"id":"7c0f08b68215fe9c01a7347d6620fc29e1d5629f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 18:22:29.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/54/786054/4/check/DellEMC_PowerMAX_FC/fbf9467/DellEMC_PowerMAX_FC/3 : SUCCESS in 5h 12m 54s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/54/786054/4/check/DellEMC_PowerMAX_iSCSI/b2b740a/DellEMC_PowerMAX_iSCSI/3 : SUCCESS in 5h 02m 16s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"662add89dd5a2e9447fe0080c90d15d9dbdd67a2","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 21:56:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/4/20210421043000/NEC-FC : SUCCESS in 2h 11m 37s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/4/20210421043000/NEC-ISCSI : SUCCESS in 2h 26m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fd58dba1ff6e466b6930717079a19093c469a891","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 22:07:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/54/786054/4/check/DellEMC_VNX_iSCSI/3c49328/DellEMC_VNX_iSCSI/7 : FAILURE in 3h 48m 12s\n- DellEMC_VNX_FC http://publiclogs.emc.com/54/786054/4/check/DellEMC_VNX_FC/5e9e374/DellEMC_VNX_FC/8 : SUCCESS in 1h 41m 26s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"f454b6b9928c9f1ff4a39d51b3632803d1f9272a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 22:56:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/54/786054/4/check/DellEMC_XtremIO_FC/6e42953/DellEMC_XtremIO_FC/19 : SUCCESS in 2h 06m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/54/786054/4/check/DellEMC_XtremIO_iSCSI/cf83836/DellEMC_XtremIO_iSCSI/19 : SUCCESS in 1h 53m 25s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"6f8d63db97d2a61a6426d26b8114c45d3f85f154","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-20 23:21:43.000000000","message":"Patch Set 4:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":4},{"id":"2cdcba29974e9363afd934fbb4bb3cc127e6e5fa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 10:05:28.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/4/check/tempest-dsvm-full-focal-py3/4b6d8b9/ : TIMED_OUT in 3h 10m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/4/check/devstack-plugin-ceph-dsvm-focal-py3/2553ae2/ : SUCCESS in 1h 38m 01s","accounts_in_message":[],"_revision_number":4},{"id":"572af57dccb485ecfcfc5ab92985dd4d27f5509b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-22 01:08:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/4/check/iscsi-driver-nimble/68e8f58/68e8f58.html : FAILURE in 1h 46m 51s","accounts_in_message":[],"_revision_number":4},{"id":"5ef0566322dcfbb9bdb722f00abdeb9f96dbcf91","author":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"date":"2021-04-22 16:29:52.000000000","message":"Patch Set 4:\n\nrun-Nimble Storage CI","accounts_in_message":[],"_revision_number":4},{"id":"412770db86bbfa2e43478cb8d937b6c29709e8b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-22 18:28:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/4/check/iscsi-driver-nimble/6aaefaa/6aaefaa.html : SUCCESS in 1h 56m 27s","accounts_in_message":[],"_revision_number":4},{"id":"9342dca3915efa05434674e5af8a5dceb94ffb3f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-30 12:40:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nSome small changes requested inline, otherwise this looks fine.","accounts_in_message":[],"_revision_number":4},{"id":"7dd57a13d92f15f25d0280bbe6bfc6240f7ef0ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-23 07:05:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c4eef98c586fd58c7a46347d1f1e6328791fc10","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-23 07:15:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"60350ad108e553ff6e8d978d595c2d094fbb5c57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 07:26:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/653c88600c9046bc8a747811b7eb7b1d : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3384915e67c6435991a8a97e64fc5d12 : SUCCESS in 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b0e42d416adeaf069125ff91f1278a7da5f24fd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-23 08:02:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/5/iscsi/ : SUCCESS in 55m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/5/smb/ : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":5},{"id":"4545ad628db010393fc1df07e0583842e22c20f0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-23 08:43:21.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-5/ns5_nfs : FAILURE in 20m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-5/ns5_iscsi : FAILURE in 20m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"e214c81adc0f9b99aee09c27354fc11c12cb4c33","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-23 08:48:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/5/check/dsvm-tempest-cisco-zonemanager-job/e6422e5/ : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":5},{"id":"b30bebdc3e9dfe56ed12614f96676607c8aeb2a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 08:52:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/54/786054/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4ec5f0c/4ec5f0c.html : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":5},{"id":"cccc651a5ebdad127442b9231d70b1f86afda0d4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 08:55:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/54/786054/5/check/iscsi-driver-primera/5a22121/5a22121.html : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":5},{"id":"124d4738e3a10ed55a50db4d4322219acc1d8306","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 08:56:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/786054/5/check/iscsi-driver-alletra-6k/93ac79c/93ac79c.html : SUCCESS in 1h 45m 39s","accounts_in_message":[],"_revision_number":5},{"id":"254c6c0175a712d3849c9ae45671b605b910c1e3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-23 09:38:50.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/65b9359/ : FAILURE in 2h 31m 44s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f8e00a9/ : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":5},{"id":"136a897aaf641315c00dc145b8720b17b5ace185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 09:42:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45872030705049a8b691557fdce06bed : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86736396bba846658401be369ae1fe48 : SUCCESS in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7990101fd014e9f98b27fe42e09079c : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d892ea562eb487d986c3aa6a7401df9 : SUCCESS in 16m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e30cb5dca54668b907f6f0fb376148 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/70b88d3eb65a4a86adcdb3a8272b61e0 : SUCCESS in 58m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4e735718684464f96e16a9844a87317 : SUCCESS in 1h 34m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1440421f8cf246de8a54d1e0ca8deb24 : SUCCESS in 9m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cda52f402f204426add50e73fd47c729 : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b7b867d5cc184a2fb7449e23d1b4668c : SUCCESS in 4m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ef51ac17d4949549f3ef15d6509d5d0 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cf216f30fe14e13acf021d9f434549c : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de5a320e82ef4cb6b4ce9352df90cbf1 : SUCCESS in 10m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4725487fb5c24bcda82c1c02ba8675ce : SUCCESS in 28m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64d797351e8c4171aae20db5cc77e4c7 : SUCCESS in 9m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e95b77a91de640bda62483757d49e6a8 : SUCCESS in 1h 11m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/32aadfa88a25492a8c2ca0ed4d9d461c : SUCCESS in 1h 24m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/357b3dea2dc14811ac2281434fd83c9a : SUCCESS in 1h 22m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9fe135ba78f74895b232fb8cf810cd18 : SUCCESS in 52m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/656d683e286c4e058f3715c061cfdb3f : SUCCESS in 1h 03m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/021154aa4e5243839f21d97d21610f7f : SUCCESS in 44m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/67168eca789943509f61afee6de12de8 : SUCCESS in 1h 37m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0429d05eba4a428483af8157c2a4d31c : SUCCESS in 2h 34m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60111c60a3fd42edbda62e926be60425 : SUCCESS in 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afb41263669f4658bf32316ce9a0c82d : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":5},{"id":"a586d32886dcf725caff9502b14a8f851cfe8bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-23 10:00:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"888057522a71ffced1668df15310b8ed61acc7a3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-23 10:01:43.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"e1765bfaa6065068383355d9ad797e12250b4420","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-23 10:07:51.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"aea3b105e60ddee2df64108360c741096d8848c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 10:22:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fbb034f95354c23a8423dd9a011c2da : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f27f376afb5b4dfbbf1c547f88aa9985 : SUCCESS in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"22c544eb3c971e3514425936ba64a29f1a0ef3e3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-23 10:57:21.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-6/ns5_nfs : FAILURE in 19m 23s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-6/ns5_iscsi : FAILURE in 19m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"2fec4b0769f681ca0a63325ecfa45b68a8678997","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-23 11:04:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/6/iscsi/ : SUCCESS in 1h 03m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/6/smb/ : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":6},{"id":"c8088620a5bb37fd96aeeecaf29772b59d13756d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 11:20:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/54/786054/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/97604e4/97604e4.html : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":6},{"id":"308d229f70d6fb6c69ae24d7afdc5901a8295de0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 11:27:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/54/786054/6/check/iscsi-driver-primera/8810787/8810787.html : FAILURE in 1h 24m 19s","accounts_in_message":[],"_revision_number":6},{"id":"df5e9a93deb38d369abd9d28848be746830f125f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 12:03:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5092610a440a460f802c18747de2021c : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f49efe726ea94798903256d606830fcd : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4c17d658e9476d83c0537dff38ff0d : SUCCESS in 11m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d609efeace3a45e7838b2ebff51065ff : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662d7aa555514b4f9d45cfc554df0414 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f7948d23965346ad962049e38ba55215 : SUCCESS in 52m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/709020f6f7d2427eaad365f7670c4264 : SUCCESS in 1h 02m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8947ab33f7c044f383ea9971da97c7a8 : SUCCESS in 7m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9d799d7fcc44ab18a247d3054245d6c : SUCCESS in 12m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7af9a78547344b21a79630d89e9bbcd2 : SUCCESS in 6m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb21f5d9e557401aa3a6bcc546ffd048 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/645480d198b8415b93e1e5865cd5c5d6 : SUCCESS in 6m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/df29fbd34c564a7ab0cc2b8c9cfa6f36 : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/390c9098b5124b9a8d7a7e13a28e348e : SUCCESS in 34m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/563ca81ea0ba471f91846336671bf765 : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a98008ce806e43f8a012f73999dff708 : SUCCESS in 1h 13m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a1edbc823914dfaa00da246f8f60fcb : SUCCESS in 1h 21m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1939592975334628a7ce7dad91571b23 : SUCCESS in 1h 25m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c50f1cbbf5184cf3a26f12494622b434 : SUCCESS in 1h 02m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c95ee97380a74faa845491c1adfb779b : SUCCESS in 1h 11m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f0c244e4af8f4401ab5c2c28519c13af : SUCCESS in 38m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6cf354e3a3f840efa17c1470d8d20491 : SUCCESS in 1h 28m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f82d9d437d564c0689e57787cd5f8702 : SUCCESS in 1h 58m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19fcd4f38070458495fbba0f28e805be : SUCCESS in 45m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dd2068064e6481689995fde7f4ff839 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":6},{"id":"82f58dc2a9e972c8f500f720982b57f553ec45dd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 12:12:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/786054/6/check/iscsi-driver-alletra-6k/b60d1e8/b60d1e8.html : SUCCESS in 1h 39m 37s","accounts_in_message":[],"_revision_number":6},{"id":"8e220640cf4b500c7826269b1a5b40dc249becf1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-23 12:56:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ee98f48/ : FAILURE in 2h 43m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/28f5262/ : FAILURE in 2h 54m 38s","accounts_in_message":[],"_revision_number":6},{"id":"cb5b89abf2a0c29f33847d12255099a68a9075b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-23 13:43:27.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/6/check/tempest-dsvm-full-focal-py3/877afd4/ : SUCCESS in 2h 15m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/786054/6/check/devstack-plugin-ceph-dsvm-focal-py3/9791bef/ : TIMED_OUT in 3h 13m 45s","accounts_in_message":[],"_revision_number":6},{"id":"14cb7b47afa2f1ec63e7c3542d6b3daf97489b2b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-23 14:04:49.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_PowerFlex/72efebd/DellEMC_PowerFlex/130 : SUCCESS in 1h 48m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"13d7625685ab0b3527c059d43e12f356e10dc6d1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-23 14:08:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/54/786054/6/upstream-solidfire-check/cinder-solidfire-iSCSI/f8b712d/ : SUCCESS in 2h 03m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"5d9a6076ac0ab5cee5d77ba6e69e025b63e1e289","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-11-23 14:47:53.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_SC_FC/63b24fd/DellEMC_SC_FC/50 : SUCCESS in 57m 51s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_SC_iSCSI/d004311/DellEMC_SC_iSCSI/50 : FAILURE in 1h 49m 38s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"dd8e26fb84dc4236ae75075708aca69db81e02e2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-23 15:04:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/54/786054/6/upstream-check/cinder-cDOT-iSCSI/5415d17/ : SUCCESS in 59m 31s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/54/786054/6/upstream-check/cinder-cDOT-NFS/a70e321/ : SUCCESS in 1h 01m 07s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/54/786054/6/upstream-check/cinder-cDOT-FCP/8787088/ : FAILURE in 2h 27m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f13fa093f71568b4b669952c83660ca42fd115af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 15:30:04.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/6/check/iscsi-driver-nimble/14fde4e/14fde4e.html : SUCCESS in 1h 49m 03s","accounts_in_message":[],"_revision_number":6},{"id":"9fe2eb5e4a0df6050a4619cec2b9c42c0340e15c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-23 17:07:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786054-5/HPMSA-iSCSI : SUCCESS in 1:11:52\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786054-5/HPMSA-FC : SUCCESS in 54:38.08","accounts_in_message":[],"_revision_number":5},{"id":"a05d8ee61263e4ffdc1348e8800ddbdfad213b08","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-23 18:17:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786054-5/Seagate-iSCSI : SUCCESS in 1:21:07\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/786054-5/Seagate-FC : SUCCESS in 49:31.02","accounts_in_message":[],"_revision_number":5},{"id":"1a6d34cfe1c40d48d5a2ddb9e2ed125095ae6cf9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-23 18:19:11.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_XtremIO_FC/9d62dd6/DellEMC_XtremIO_FC/205 : SUCCESS in 1h 10m 32s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_XtremIO_iSCSI/3f40403/DellEMC_XtremIO_iSCSI/223 : FAILURE in 1h 57m 22s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"b9a2aa35efcd41d0407d3a674cfb11c590dfb7d8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-23 18:58:57.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_PowerStore_iSCSI/578c19d/DellEMC_PowerStore_iSCSI/107 : SUCCESS in 2h 07m 39s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_PowerStore_FC/11e103b/DellEMC_PowerStore_FC/118 : SUCCESS in 1h 25m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"6d80d76d45ae2ebc68873d0bea7f3b35947a2dbf","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-23 21:07:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786054-6/HPMSA-iSCSI : SUCCESS in 1:11:26\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786054-6/HPMSA-FC : SUCCESS in 43:10.81","accounts_in_message":[],"_revision_number":6},{"id":"e6f7f03821d1932cf2e3d2bf5e1c0a6894b20493","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-23 21:19:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786054-5/Lenovo-iSCSI : SUCCESS in 1:32:18\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786054-5/Lenovo-FC : SUCCESS in 57:18.47","accounts_in_message":[],"_revision_number":5},{"id":"bee21a144eaeb5a9b5bee0f5fa1eb1caf089a778","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-11-24 03:19:07.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_PowerMAX_FC/a993ac9/DellEMC_PowerMAX_FC/62 : SUCCESS in 3h 09m 36s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_PowerMAX_iSCSI/4d1c579/DellEMC_PowerMAX_iSCSI/61 : FAILURE in 3h 08m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"4626a644c71d0d1c753617ff181b272b89fd981e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-24 03:26:33.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_VNX_iSCSI/2cdd78f/DellEMC_VNX_iSCSI/117 : FAILURE in 3h 04m 44s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/54/786054/6/check/DellEMC_VNX_FC/6e0f374/DellEMC_VNX_FC/118 : FAILURE in 2h 53m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"48aa981547a485223b64eb70b0b67f22a98f72e7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-24 06:34:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/6/20211124124341/NEC-ISCSI : SUCCESS in 2h 41m 17s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/6/20211124124341/NEC-FC : SUCCESS in 2h 50m 41s","accounts_in_message":[],"_revision_number":6},{"id":"4f34eada6fe9094437090cc31a513891f088022e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-24 06:36:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"199e4a3f6cf122e2c295661f59620d05284fe8cd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-24 12:31:50.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786054/786054/6/check-not-vote/ext-cinder-zuul/f7ec56f : SUCCESS in 32m 21s","accounts_in_message":[],"_revision_number":6},{"id":"d96eb5f2f34c5a5a57d99d23e75e7c1e0ef8cb15","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-24 17:49:08.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-6/ns5_nfs : SUCCESS in 4h 14m 13s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-6/ns5_iscsi : SUCCESS in 4h 19m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"8dc07bc669de229f439daa948447a831d951d690","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-24 17:56:37.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c983ecb069b82f82ee2b941a3251ad343204db01","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-25 10:19:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/54/786054/6/check/dsvm-tempest-TOYOU-ACS5000-driver/d461f6c : SUCCESS in 2h 05m 31s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/54/786054/6/check/fc-test/5d91666 : SUCCESS in 1h 55m 54s","accounts_in_message":[],"_revision_number":6},{"id":"71acf3ad929da90de0719d2d1305fd8432127031","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-29 12:25:41.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6bbbebc9f1050098c8672a826242a04c8c25cd8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-30 13:56:48.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b95d81fc43c020ac5fe83b81a405ad7eeae26646","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-11-30 17:04:31.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d3db47e513c4eadb5c7d0268e20ba401be022886","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-12-01 12:55:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4a3d74884b23834d7a2845afd632b8401a1c1263","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-01 14:45:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/6/check/iscsi-driver-nimble/bcbbbcc/bcbbbcc.html : SUCCESS in 1h 48m 35s","accounts_in_message":[],"_revision_number":6},{"id":"87047c1cd3071848a72d0bd8001f485c2a2ce43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-12-02 07:36:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"232afb316e33371eeb9e864f15a98f66867e037d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-02 07:41:07.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/4e0cdc8a211d4e8dbb5e8607d9442edc : RETRY_LIMIT in 18s","accounts_in_message":[],"_revision_number":7},{"id":"9dd5e99dfb9b9519ff15e63068004d0359f67bf7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-02 07:50:52.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786054/7/iscsi/ : FAILURE in 13m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/786054/7/smb/ : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":7},{"id":"5be10ae9d56b1c687276da2f7a180a88e738087b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 08:00:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3970d31df624645a4c850c8e4fb65f0 : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ac315d226ed44c9bb155d03d9d3de22 : SUCCESS in 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ef516b3d45a245bf51a691eea85bc55a91588123","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-02 08:02:39.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"a1628ccb2685df4bb37e2e09be14b6eb8a0e4854","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-12-02 09:18:03.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b66183ccb06c81380802cb065e8462a5cda78e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 09:47:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6694a6740ca4834ab5d609386ed808d : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3613c256fd9041dc9677800dec555812 : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adea921a2a694407a761dd52ca97c869 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc558cc4cc9949b4ae8437c7fe1265a4 : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/db163c91d2ea411384bd96125aaaeb55 : SUCCESS in 1h 17m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/572e57c70a67433286c78cde0832c6dc : SUCCESS in 1h 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83795a842a694f8db8f0b7d1f1af84d8 : SUCCESS in 6m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/63a2ec190e5a4238911e645dad821597 : SUCCESS in 14m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/39ddd46cca6d4ad48cb2938bd34bbd1c : SUCCESS in 5m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/464e6b86aff742bf86c4726b89b0bed6 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4f3ac2914824c5dac8c47f60040bbf4 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/948d50215130497a875abdace8612afe : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2298822164294632aa2d08a2a6362a86 : SUCCESS in 40m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e48987f6bb934d50ad2596b9eafd780c : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3f70620740c4ded9d3b1aa455607c5b : SUCCESS in 1h 20m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/68fc67a0bf224028877f50b9249c19b9 : SUCCESS in 1h 34m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/130d8ce676be440a8f048371a52c9f85 : SUCCESS in 1h 40m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/95e7e5515cde479da4def275208bd1a8 : SUCCESS in 54m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48e1ab1d577a4bc7a47988f8ab05234e : SUCCESS in 1h 08m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cb0239ff161a4cd4af1350233b8fd1e5 : SUCCESS in 34m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c71a3537457d450fb66b6fac317ec782 : SUCCESS in 1h 17m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/301f7947eb094261b816e86642be36aa : SUCCESS in 2h 08m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30659be399824e2bb901bd0d819ea5cb : SUCCESS in 42m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84acb7365db047efbc2935d47f7fcdb8 : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":7},{"id":"631365fc13f7f496fefedeb2a471e02eabd627d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-12-02 10:01:24.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_PowerFlex/573f4b5/DellEMC_PowerFlex/173 : FAILURE in 13m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"aa29f763ab473498e034594c4d831935d731c0b8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-02 10:17:58.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_Unity_FC/d394642/DellEMC_Unity_FC/247 : SUCCESS in 1h 20m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_Unity_iSCSI/5cf516f/DellEMC_Unity_iSCSI/249 : FAILURE in 5m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"17fcc31ad5059be4b3fb621765a9694d2b73c939","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-02 10:21:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/54/786054/7/check/dsvm-tempest-cisco-zonemanager-job/9a08b72/ : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":7},{"id":"635c2a13115be1b785c664588ff76f568e02ab91","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-02 10:22:19.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/54/786054/7/upstream-solidfire-check/cinder-solidfire-iSCSI/1104621/ : FAILURE in 33m 50s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"8004417d2bfa6e0169b5f250f4f4ff9b348680cb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-02 11:20:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/857019b/857019b.html : SUCCESS in 1h 48m 19s","accounts_in_message":[],"_revision_number":7},{"id":"8a826e06f352e19a58a444193c7d5ffe90290019","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-02 12:17:03.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_XtremIO_FC/ec71155/DellEMC_XtremIO_FC/280 : SUCCESS in 1h 01m 14s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_XtremIO_iSCSI/07bdf1f/DellEMC_XtremIO_iSCSI/295 : SUCCESS in 1h 34m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"300bcae941b951343be689198acc209d5a527903","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-12-02 12:25:03.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_PowerStore_iSCSI/9dd6d64/DellEMC_PowerStore_iSCSI/142 : SUCCESS in 1h 35m 15s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_PowerStore_FC/1f807de/DellEMC_PowerStore_FC/153 : SUCCESS in 1h 03m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"e23cb888a98fbf5d970d4394883a97cac6b63cf6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-02 12:40:38.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_SC_FC/56a6bdc/DellEMC_SC_FC/96 : SUCCESS in 50m 43s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_SC_iSCSI/418500a/DellEMC_SC_iSCSI/95 : SUCCESS in 1h 43m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"67f325b7919c1f67d4ae17f96b52acf5d83abccc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-02 13:12:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/54/786054/7/upstream-check/cinder-cDOT-iSCSI/0ecbf5e/ : SUCCESS in 2h 46m 22s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/54/786054/7/upstream-check/cinder-cDOT-NFS/afeb225/ : SUCCESS in 2h 43m 05s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/54/786054/7/upstream-check/cinder-cDOT-FCP/2e73f39/ : FAILURE in 3h 24m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"de79ce2b4dec7ca41fd22c90e1f6e5774164a852","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-02 13:26:03.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-7/ns5_nfs : SUCCESS in 4h 27m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-786054-7/ns5_iscsi : SUCCESS in 4h 31m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"565800a28eb5cde4fa25623d61eb92e362e1c710","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-12-02 14:02:45.000000000","message":"Patch Set 7:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/159/index.html","accounts_in_message":[],"_revision_number":7},{"id":"d03a44c47745edfa642661b14a697c0fe0377bd7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-02 14:04:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786054-7/HPMSA-iSCSI : SUCCESS in 1:11:50\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786054-7/HPMSA-FC : SUCCESS in 44:13.08","accounts_in_message":[],"_revision_number":7},{"id":"f0d9eecf32b7a6428400c2b93a1150289455fad0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-02 14:29:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a257923f1d3d40c3a991eb0752576b25 : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":7},{"id":"8f958e3ceac4aad67124ce835e3c838fd06d6f10","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-02 20:30:48.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/786054/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0f37985/ : FAILURE in 2h 37m 04s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/786054/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7ec3226/ : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":7},{"id":"660ae8d3929cc87c6c27e02f0424b1b9be34bc93","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-12-02 22:20:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786054-7/Lenovo-iSCSI : SUCCESS in 1:28:11\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786054-7/Lenovo-FC : SUCCESS in 57:50.42","accounts_in_message":[],"_revision_number":7},{"id":"e8c26436e9337e1f3d15f7653a0a07e2a020a421","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-12-03 06:19:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7ba12be349c52f8570066a01146c1914a846e869","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-12-03 09:46:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"83cbd243573f6f0988e0581b54f96174612026d0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-03 12:55:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/28187aa/28187aa.html : SUCCESS in 1h 49m 22s","accounts_in_message":[],"_revision_number":7},{"id":"d05883b378bc42eace18a12927b6e25174c09c55","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-12-03 16:03:23.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_PowerMAX_FC/3ea55ff/DellEMC_PowerMAX_FC/104 : SUCCESS in 3h 31m 23s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/54/786054/7/check/DellEMC_PowerMAX_iSCSI/942cda2/DellEMC_PowerMAX_iSCSI/102 : SUCCESS in 3h 41m 51s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"383f0deea97761e9c4b0b9a34db19e1bae949540","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-04 00:38:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/7/20211204064519/NEC-ISCSI : SUCCESS in 2h 45m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/786054/7/20211204064519/NEC-FC : SUCCESS in 2h 53m 11s","accounts_in_message":[],"_revision_number":7},{"id":"ca2b4240e4bce7fb61d8e332dbd77eca240b1464","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-04 01:19:13.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/54/786054/7/check/dsvm-tempest-TOYOU-ACS5000-driver/3f44f42 : FAILURE in 5h 00m 24s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/54/786054/7/check/fc-test/7c9fc90 : SUCCESS in 2h 12m 34s","accounts_in_message":[],"_revision_number":7},{"id":"a4c7efc7e708ffb4330420ffda23cf3f63b48669","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-06 07:44:23.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/786054/786054/7/check-not-vote/ext-cinder-zuul/b3846cb : FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"41cf0edd31143ed8e79d846c0af80335ef5d16c7","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-12-07 11:51:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d4e43b1a5cc8155769c6d88cd77a057b78949320","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-01-20 10:44:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"384cb15617b4935b685f50dc15c5048cbe7ee840","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-20 11:48:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/54/786054/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/67d095f/67d095f.html : FAILURE in 1h 02m 39s","accounts_in_message":[],"_revision_number":7},{"id":"24b0943735a3d76f4713b2f415507830b64f2c0c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-01-31 09:05:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a00538cbe091dc08ca015138b2f0ea8e95dbc534","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 10:27:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/54/786054/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/69f4621/69f4621.html : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":7},{"id":"0690664ab638dc72ac5ae66302d508cefb41dec3","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-07 13:21:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8d99d8bd5649fbb0aad3fb1feb8a26485ebc4944","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 14:03:09.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6436ee9d82dba4a30bb9116f45ca10d22e824d6a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-08 06:28:24.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"caaefea538013a0f7fb3c584b59967d5643e4cd0","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-10 13:13:05.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b28c5c735dadef3995649d01f5e06a3cb1255448","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-10 13:13:42.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef40f6c6a94d97e3b23b8465f60ace2680bf0568","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-10 14:35:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"85c33e6493322923191082e9fc4412d597c83d98","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-16 09:31:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6f80dc098048dd53918fe25dc479a585b4e4e194","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-17 23:43:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"80a0f9a8def93fa56a33720c23265416abc07233","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-18 12:22:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"68d82106c4e53a29e736568b94a6a6fc6fc64f9d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-18 13:17:48.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2c7a2fbea04d500d68023ee949128a7b6dc5a77d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-21 10:40:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9d5a5466c7007de3f74ffd5cc59893afa5984a76","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-21 12:06:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0067cad42434cf3b864b2bc341c747cdef7e6ed8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-21 12:40:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e4f1d5bd1b0cf6e6c2de2a3349fc2e74bb4b6fff","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-21 16:22:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/62bda8d/62bda8d.html : SUCCESS in 1h 49m 27s","accounts_in_message":[],"_revision_number":7},{"id":"b0150f45963d8c74e65c8c98b9f83fd4c989c33b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-22 00:18:26.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/bc444d8/bc444d8.html : TIMED_OUT in 3h 00m 45s","accounts_in_message":[],"_revision_number":7},{"id":"6aa806b567eb7989e66eb976242d33226579f110","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-22 05:37:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"36a372fe4cd62726da4bce82a9de19524c920614","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-22 05:38:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d4b261f56ff35243ad52c445061b82a4542583d8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-22 14:31:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/7f76468/7f76468.html : SUCCESS in 1h 50m 45s","accounts_in_message":[],"_revision_number":7},{"id":"8cd33ffd802199ca55b8c87601196323f4777f41","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-22 16:01:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8ccf2250d2aa74ea9b138eaaf8bb3882c7057187","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-22 16:23:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/1fd6b72/1fd6b72.html : SUCCESS in 1h 50m 54s","accounts_in_message":[],"_revision_number":7},{"id":"a90ba4a0830e9ac0a8548d6e8a13d740ac726ad0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-22 16:51:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8986c597eb3baac5c078f40964b05bea3192230","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-23 06:01:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e6fe7c5f4fba54974963377c65292d83666c3ffc","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-23 13:40:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f03d1489d1230fb7de5d7aabe01d49b4cb0f1062","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-24 03:23:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/7705fda/7705fda.html : FAILURE in 1h 49m 51s","accounts_in_message":[],"_revision_number":7},{"id":"94575476a9cac71f83c6dd3acf48705cf5c21f22","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-24 05:08:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4974b5f573b6a5792f725b587a2d6a3220241d0f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-24 05:35:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/69d3260/69d3260.html : FAILURE in 2h 10m 35s","accounts_in_message":[],"_revision_number":7},{"id":"96635361eb0680bd62c509cbffee6f5c10f471b5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-24 06:12:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"3b71aa3f431a2cffa57b0ab35a9916cb5eb693df","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-24 23:23:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f202ec2b1b4723ae89ffb8d631752efff19c9b11","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-25 05:21:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"132afd8ea7eba903dd7b00a039dc8da457f20d65","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-25 07:04:53.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/e46dda9/e46dda9.html : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":7},{"id":"cd016960c7b18957121072078be9e385e037053b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-25 09:21:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/786054/7/check/iscsi-driver-nimble/6de32e7/6de32e7.html : SUCCESS in 2h 14m 42s","accounts_in_message":[],"_revision_number":7},{"id":"10114453740e24d25f23206e21089e22c9fcb59f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-25 10:31:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2cc3fce2e3ff4c9ed119cabf9b52b374762a89e2","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-25 10:34:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"58112ac7c3ed6611375cf5fa09d2f2e9d453c7ef","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-25 13:30:32.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"89b576a275216fd1e74427f46b9bdc06605b4be3","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-02-28 12:25:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0012eb99dd14da2785bf9366204b0e15e7621642","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-02 10:31:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"656a17fb5162a8e76826d945368f64e360434843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:32:56.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7acc6a9639be94c9ccba3465c64cc33cf4b79709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:54:19.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38d451c20bd44128505672e48d61edd : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f86e25ad0e845298fce34cfc2945d85 : SUCCESS in 12m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05b2291acb724aea9ea7a329bf9c2430 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71bb02556fb46bea3c8ebb1ea041096 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/18ff8db702574942a62bec0289b5e8f1 : SUCCESS in 42m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/894dbcea45c24d63b0142cc149889b27 : SUCCESS in 1h 17m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1ff79d35600411c9c2e36e6444b5c45 : SUCCESS in 39m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eb15ae8ca344678b21bef380bc4a437 : SUCCESS in 8m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbc76cabe166412bbe19e839207f038e : SUCCESS in 1h 06m 03s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d8e61fc282c44683a5dc931462ef511e : SUCCESS in 1h 12m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea8ac11c1890421792a1ff91771a5320 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":7},{"id":"9e6a6fe993d91fdb63be7f39fd91a64d6f4a880d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:54:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"165083e81909878481f180188c7ae2951bc774d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:56:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e43d3006a48419bac8b5f55d21f51c3 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6aac078381844c3b31676dac15359e6 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":7},{"id":"56380d2c383bc5a6e87c68cfe55672f1e06fecc0","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-03-02 12:26:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d67dcf42d4cc7c727db115b1e0a9c8a8a7270b2b","revisions":{"30381238b6d4d1d8e0782d9e28c040fb9dfd9083":{"kind":"REWORK","_number":1,"created":"2021-04-13 13:53:34.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/54/786054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/1"}}},"commit":{"parents":[{"commit":"c12375fec6c005ba8d7e14d6f94612e7f5e66dd2","subject":"Merge \"Backup manager: Synchronously call remove_export_snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12375fec6c005ba8d7e14d6f94612e7f5e66dd2"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:53:32.000000000","tz":-420},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30381238b6d4d1d8e0782d9e28c040fb9dfd9083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30381238b6d4d1d8e0782d9e28c040fb9dfd9083"}]},"branch":"refs/heads/master"},"9eafb7f53d45dd0f903da83607462bbfc89e7e87":{"kind":"REWORK","_number":2,"created":"2021-04-14 06:20:40.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/54/786054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/2"}}},"commit":{"parents":[{"commit":"c12375fec6c005ba8d7e14d6f94612e7f5e66dd2","subject":"Merge \"Backup manager: Synchronously call remove_export_snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12375fec6c005ba8d7e14d6f94612e7f5e66dd2"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-14 06:20:15.000000000","tz":-420},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9eafb7f53d45dd0f903da83607462bbfc89e7e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9eafb7f53d45dd0f903da83607462bbfc89e7e87"}]},"branch":"refs/heads/master"},"c638f49374b07a2543823aad3da0c1bbda559049":{"kind":"REWORK","_number":3,"created":"2021-04-19 02:51:39.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/54/786054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/3"}}},"commit":{"parents":[{"commit":"c12375fec6c005ba8d7e14d6f94612e7f5e66dd2","subject":"Merge \"Backup manager: Synchronously call remove_export_snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12375fec6c005ba8d7e14d6f94612e7f5e66dd2"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-19 02:51:18.000000000","tz":-420},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nblueprint nimble-change-location\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c638f49374b07a2543823aad3da0c1bbda559049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c638f49374b07a2543823aad3da0c1bbda559049"}]},"branch":"refs/heads/master"},"579214e7f797d8506e5707563a286b10093d0e66":{"kind":"REWORK","_number":4,"created":"2021-04-20 05:03:53.000000000","uploader":{"_account_id":16186,"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","username":"ajitharobert"},"ref":"refs/changes/54/786054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/4"}}},"commit":{"parents":[{"commit":"c12375fec6c005ba8d7e14d6f94612e7f5e66dd2","subject":"Merge \"Backup manager: Synchronously call remove_export_snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12375fec6c005ba8d7e14d6f94612e7f5e66dd2"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-20 05:03:30.000000000","tz":-420},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nblueprint nimble-change-location\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/579214e7f797d8506e5707563a286b10093d0e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/579214e7f797d8506e5707563a286b10093d0e66"}]},"branch":"refs/heads/master"},"44d44954502ecee947fe01d90d7b37b61b3af98c":{"kind":"REWORK","_number":5,"created":"2021-11-23 07:05:57.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/54/786054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/5"}}},"commit":{"parents":[{"commit":"2969504e98ac503d1f1e39353ea377b09c7390ab","subject":"Merge \"pylint: Update to 2.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2969504e98ac503d1f1e39353ea377b09c7390ab"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2021-11-23 07:14:36.000000000","tz":0},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nblueprint nimble-change-location\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44d44954502ecee947fe01d90d7b37b61b3af98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44d44954502ecee947fe01d90d7b37b61b3af98c"}]},"branch":"refs/heads/master"},"2bd31a72a6efce0c7f4dde530c9257331d04cd67":{"kind":"REWORK","_number":6,"created":"2021-11-23 10:00:58.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/54/786054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/6"}}},"commit":{"parents":[{"commit":"2969504e98ac503d1f1e39353ea377b09c7390ab","subject":"Merge \"pylint: Update to 2.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2969504e98ac503d1f1e39353ea377b09c7390ab"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2021-11-23 10:08:30.000000000","tz":0},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nblueprint nimble-change-location\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bd31a72a6efce0c7f4dde530c9257331d04cd67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bd31a72a6efce0c7f4dde530c9257331d04cd67"}]},"branch":"refs/heads/master"},"d67dcf42d4cc7c727db115b1e0a9c8a8a7270b2b":{"kind":"REWORK","_number":7,"created":"2021-12-02 07:36:32.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/54/786054/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/786054/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/786054/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/786054/7"}}},"commit":{"parents":[{"commit":"2969504e98ac503d1f1e39353ea377b09c7390ab","subject":"Merge \"pylint: Update to 2.11.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2969504e98ac503d1f1e39353ea377b09c7390ab"}]}],"author":{"name":"Ajitha Robert","email":"ajitharobert01@gmail.com","date":"2021-04-13 13:44:12.000000000","tz":-420},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2021-12-02 07:43:44.000000000","tz":0},"subject":"Move nimble driver code to hpe folder","message":"Move nimble driver code to hpe folder\n\nblueprint nimble-change-location\n\nChange-Id: Id1f111462fc2558b90f9066387498bcb5e3217b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67dcf42d4cc7c727db115b1e0a9c8a8a7270b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67dcf42d4cc7c727db115b1e0a9c8a8a7270b2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
