)]}'
{"id":"openstack%2Fcinder~786260","triplet_id":"openstack%2Fcinder~master~Iec111ab83e9ed8182c9679c911e3d90927d5a7c3","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Iec111ab83e9ed8182c9679c911e3d90927d5a7c3","subject":"RBD: use correct stripe unit in clone operation","status":"MERGED","created":"2021-04-14 15:15:55.000000000","updated":"2021-08-09 16:23:21.000000000","submitted":"2021-08-09 16:21:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"786260","meta_rev_id":"317ca0ddf13f4c4a1adcc303fff2cca8ccca206e","_number":786260,"virtual_id_number":786260,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-09 16:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-08-04 12:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-09 15:01:28.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-09 15:01:28.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-08-04 12:51:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 15:18:08.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 15:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 16:22:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 16:22:57.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-14 16:40:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-14 17:11:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-14 17:20:09.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 17:40:05.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 17:54:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 19:57:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-15 07:43:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 14:34:23.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-05-03 20:20:29.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-06-06 11:44:46.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-06 12:17:43.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-06 14:51:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-07 10:04:08.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 11:56:05.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-06-08 21:05:31.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 20:06:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 20:14:10.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-06-17 20:50:17.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-06-20 23:40:28.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-06-20 23:40:32.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-06-21 03:45:23.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2021-06-21 11:12:09.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-06-30 00:18:26.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-30 01:37:34.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-06-30 10:35:59.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-06-30 14:03:03.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 13:58:44.000000000","updated_by":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"reviewer":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"state":"REVIEWER"},{"updated":"2021-07-22 14:03:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-07-27 22:13:30.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-07-28 05:57:18.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-07-29 13:05:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-02 05:33:02.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-08-03 22:32:40.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"}],"messages":[{"id":"8d41f1423fd1f157f184d0e86350b2fedc12f5c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-04-14 15:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f73ec2bb5d58a53244400c1638965ff08a2c083e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-04-14 15:17:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5662e84d7d8bdf7d123e5a47d58f85fa1e32b999","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 15:18:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"6f9a7e3b7447d1b812355e9891940a8a42ce34d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:48:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ef6037e28fc448e81faa96f74af8dbc : FAILURE in 19m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8966503d332b46fa95b76cadb4eb2d9c : FAILURE in 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05b6c585e597c93d54f5deaf14ad0b9808c3d344","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 16:22:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/786260/2/check/iscsi-driver-nimble/5a0c5a5/5a0c5a5.html : FAILURE in 1h 00m 08s","accounts_in_message":[],"_revision_number":2},{"id":"20d52b78529d43346568ec01c2a8292aeb5f5da3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-14 16:22:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/2/iscsi/ : SUCCESS in 1h 04m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/2/smb/ : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f64e829679b7b832f9e1c93d860b5569e66cf9a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 16:26:36.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"adabd8a37f291e8f6ed1352b5e831405fdc86b6f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-14 16:40:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/2/check/dsvm-tempest-cisco-zonemanager-job/a8a0025/ : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":2},{"id":"37bb7585e6aa77207a763da045e680bdd3b822a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 16:53:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/2/check/iscsi-driver-primera/3dcae9b/3dcae9b.html : SUCCESS in 1h 34m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4baf850388de6e3764de0807cfa7fdc3b7d5f09c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-14 17:11:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c06e582/ : SUCCESS in 1h 51m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d5b6267/ : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f3161c6eb818942e926a1132b998efb03387633f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-14 17:20:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/2/check/DellEMC_XtremIO_FC/64ef1d8/DellEMC_XtremIO_FC/1305 : SUCCESS in 1h 57m 58s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/2/check/DellEMC_XtremIO_iSCSI/aacb6fb/DellEMC_XtremIO_iSCSI/1328 : FAILURE in 3m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"655768c0146609928f788370f8de661cddc6bf0e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-14 17:40:05.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fb462dfbd4bc4530b5e1634abe27ce85 : RETRY_LIMIT in 27m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/2/check/devstack-plugin-ceph-dsvm-focal-py3/d1ae755/ : SUCCESS in 1h 43m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6c8c027872abb66a16a73e5015c1ad7dfac1b2ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:54:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2be0ea35f01149b7a26ba9373e7e08d8 : FAILURE in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60905f2336b440ce94b2304f5e8c3150 : SUCCESS in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/537aed700950428d9f734281874b3e23 : FAILURE in 15m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8f1e96c2bb4a2eab16e3a7f19abfa9 : FAILURE in 17m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d25200fab3ba403ca3ecadec8a3378e7 : FAILURE in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365cf02b70b2401ba88b4a65f372093e : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/1e58e797e2824e5497ba30abe8f31c30 : SUCCESS in 1h 03m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01d6b4b52ce041f983f217236312d2b4 : SUCCESS in 1h 15m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e1309dea0f34de79817a2941a88c18b : FAILURE in 17m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8893dc2ade5c495fa22c3e45c9b5b5a3 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8cdb3611c5794f32bb497a08d954cefd : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76da8c8bf2634400954882b5e258348b : SUCCESS in 8m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8be1ae2d284440c7b29688e7a8ec5f08 : SUCCESS in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f1ec443dd92040dfac835d760859b576 : FAILURE in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d5eb6ed64b64cb2b328965ea9091604 : SUCCESS in 1h 54m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b5c02ac5c8c46eda80a3712392df306 : FAILURE in 1h 34m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c9e7e720090461db1b3584ba4137cef : SUCCESS in 1h 28m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8c13d2410e44b1f953a58a5592724a6 : SUCCESS in 1h 31m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/855260e5975e419987f2e10ad1db8d24 : SUCCESS in 1h 22m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba2349d2f00e482d90e9f8a7beef7f8b : SUCCESS in 1h 22m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6cbfe82976df4f489fb3786a3542fa5b : SUCCESS in 2h 00m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cbd3d3e8d9b4e0586039a7a32e95f98 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7a4cad740873d9c13bbce8f1ab695483e65625c4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-14 19:57:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-1/ns5_nfs : SUCCESS in 4h 40m 46s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-1/ns5_iscsi : SUCCESS in 4h 37m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"0151f816c34e597d712fff38fcb9ff68e2a853d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 19:58:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"b2c7dadbf56379a72124c690d7b0af2f4b0d9ea7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-14 20:44:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-2/ns5_nfs : SUCCESS in 4h 41m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-2/ns5_iscsi : SUCCESS in 4h 42m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"6ff4d57715f442ecbcd4ad2bbfe15637328fd52f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 20:45:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"cbcd1f8f6e2b834401dc1d279712fde9c4437954","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 22:12:14.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c5bc57336a92b0b3a5de4e45a2b2794b35dd4cbe","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-15 07:43:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/2/20210415142332/NEC-ISCSI : SUCCESS in 2h 11m 10s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/2/20210415142332/NEC-FC : SUCCESS in 2h 19m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d322736230af1303e75ff8f1496b925cb2db7ab4","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-29 14:34:23.000000000","message":"Patch Set 2:\n\nThere are a bunch of\n\n TypeError: \u0027\u003e\u0027 not supported between instances of \u0027MagicMock\u0027 and \u0027int\u0027\n\nin cinder.tests.unit.volume.drivers.test_rbd.RBDTestCase","accounts_in_message":[],"_revision_number":2},{"id":"61fcb3873f7f52f0c1b7d0592a798a4f910aed06","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-06-06 11:11:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bda50d912509de0287fc1ecc8fd4d1b924876c56","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-06 11:34:25.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"89dd0b5262d4641f06c58a6c911b71f6e0c215d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-06 11:34:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b8aa61fff884026b0ebd11bb995ff53 : FAILURE in 12m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6514eca72fd474091edc988d0bca914 : FAILURE in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c4767cb5131001a981b165c2bc44300251773c2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-06 11:44:46.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/3/check-not-vote/ext-cinder-zuul/b177bfb : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":3},{"id":"09c94bfbfdf0a85f9ee8710eb46c4db62fd2983f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-06 11:47:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/3/check/iscsi-driver-primera/6e337be/6e337be.html : FAILURE in 24m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e5461595cfa23796697d8dc60d1538f05f5fe369","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-06 12:05:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/3/iscsi/ : SUCCESS in 50m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/3/smb/ : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":3},{"id":"cf1c39c2d605519f55cdf3a9121b2edbab3c6c5c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-06 12:17:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":3},{"id":"8ee1b3092916ea8776caac47275bb75f5abc617f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-06 13:14:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/3/check/dsvm-tempest-cisco-zonemanager-job/1182ae3/ : FAILURE in 56m 08s","accounts_in_message":[],"_revision_number":3},{"id":"524d56f93d0d518c72cfb6c60cd89031240d8ed6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-06 13:22:35.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/3/check/tempest-dsvm-full-focal-py3/db5dffc/ : SUCCESS in 2h 05m 28s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/3/check/devstack-plugin-ceph-dsvm-focal-py3/a4fd3db/ : FAILURE in 40m 23s","accounts_in_message":[],"_revision_number":3},{"id":"c1aa4c4db8522d5aa17e3e767eba8c70e4f1ae17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-06 13:23:13.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"ad213e7c1db841ff6cd511ca7fdb949d874bb54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-06 13:36:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aaba3e3f50d4badbbea9c9e79e97aea : FAILURE in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44dacdb56bc94cbf91ae1ec970ae6a02 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda3306464ed4a77af20a9d710e747dc : FAILURE in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22542bb568fb461db20bf42558317ee9 : FAILURE in 10m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87bb3d423c8f44c5b3056a3d3ab3108a : FAILURE in 11m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981e5b926e7d4cfa8360236aeb315fcc : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/306344dfaf614a50b3db1e79b25d050b : SUCCESS in 1h 07m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e189175c96fc437283fdd189aef30cb6 : SUCCESS in 1h 15m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/78103455fd784a2aa3b677b6f1b1487a : FAILURE in 12m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b64ae798d0ab4b3a84eabd8b01d3b351 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0da92accfbfb4724b70cef622a3e7ccd : SUCCESS in 8m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ddac7847a184da1986b06e823579439 : SUCCESS in 7m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/56bfa04f46c64d82bc954d6d966efc7b : FAILURE in 39m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9808b05e779b4c3f812ea95249a7e211 : FAILURE in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e8c06987dd34cf58c2766615b19d4ac : SUCCESS in 1h 24m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/499f91e358164906942092204a4d92f3 : SUCCESS in 1h 41m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f16b31e151a14946846368477e10bcd3 : SUCCESS in 1h 31m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99f906d17bf04862aded1731da56bcaa : SUCCESS in 1h 04m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f257e24f3d540f08f6738f14e6e1cdb : SUCCESS in 1h 27m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/88560df17e7442e4be0f5e3e6ce8ffcb : SUCCESS in 1h 24m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76b14994a4894ee79a49eacdf5ca83d1 : SUCCESS in 2h 20m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be44c0edbd9e4a32ba1be508a1a842df : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":3},{"id":"914ca66319299429ca65b34b0317ba9009b31b47","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-06 14:51:16.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/3/check/DellEMC_Unity_FC/3a6aa31/DellEMC_Unity_FC/236 : SUCCESS in 1h 25m 44s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/3/check/DellEMC_Unity_iSCSI/e6e5279/DellEMC_Unity_iSCSI/222 : SUCCESS in 1h 43m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"49dae41079a6daa83d1bb4e3937a453c3c7bf515","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-06 14:52:56.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"fbb681a5837b99fb28ca85bc5af00dbfd29a7655","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-06 15:19:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/3/check/DellEMC_XtremIO_FC/7a9aca9/DellEMC_XtremIO_FC/225 : FAILURE in 1h 15m 16s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/3/check/DellEMC_XtremIO_iSCSI/e32a6f9/DellEMC_XtremIO_iSCSI/230 : SUCCESS in 1h 53m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"263c41ef3b102eb24168f6573ce16d345ee3e294","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-06 15:19:48.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"f1ef1e165e91869197e13f985e38f2a5c2ed40d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-06 16:27:36.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-3/ns5_nfs : SUCCESS in 4h 27m 7s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-3/ns5_iscsi : SUCCESS in 4h 30m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"002b1aee9c4cd00b5f5f98da95f9c3d0e9cb36be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-06 16:28:32.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"07be5e0d65b45be40d289a20273c9eb53a87bd62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-06 17:56:39.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c6e360e90de98b2b29aa2134a85ee697bf10c904","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-06 19:58:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/3/20210607011058/NEC-FC : SUCCESS in 2h 24m 10s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/3/20210607011058/NEC-ISCSI : SUCCESS in 3h 47m 25s","accounts_in_message":[],"_revision_number":3},{"id":"d11c7ba8b90615507445db76dbba3f677f5b6eb5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 10:04:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/60/786260/3/check/DellEMC_PowerMAX_FC/add4b79/DellEMC_PowerMAX_FC/119 : FAILURE in 9h 56m 04s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/60/786260/3/check/DellEMC_PowerMAX_iSCSI/b280c73/DellEMC_PowerMAX_iSCSI/106 : FAILURE in 6h 40m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"1cac3cb812d7395ad31e74c8da195092b9024ed9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 11:20:56.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/786260/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9009ad8058b37a7b1d5be7c074ac8d01995ef1a1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-07 11:56:05.000000000","message":"Patch Set 3: Code-Review-1\n\nit looks like the unit tests for the affected file need to be fixed as well","accounts_in_message":[],"_revision_number":3},{"id":"cd13ca50e066ade24455e7b6f3e949ec673b995f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-08 21:05:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/786260/3/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":3},{"id":"a307dd16e989ae289923bcb728f6a160ce604672","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-06-10 18:07:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1e4e8921fa3d46841639691c0e9e1380a0c3b2c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-10 18:09:00.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/4/check-not-vote/ext-cinder-zuul/19fcbf4 : FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"66bb7937a63c328a5598cccf7d076ed31a72f1fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:31:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f210754693e408ba1aeea634dd98ab9 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fc0bb2606144144a2c0f9dd927af2fa : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8ea10ba33a65221b9479f8b435b59af3a3a3f28","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-10 18:40:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":4},{"id":"c448bac9cf92b840ffcd042a343abe266d23c0a4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-10 18:59:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/4/check/iscsi-driver-primera/71710db/71710db.html : FAILURE in 49m 54s","accounts_in_message":[],"_revision_number":4},{"id":"4253ac7a67f270e4b8ffe36af3b584ff955ff27e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-10 19:01:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e3c6c3c/e3c6c3c.html : FAILURE in 45m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c572f269a568f6bcbdfee9a706caae9c860015ad","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-10 19:11:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/4/iscsi/ : SUCCESS in 1h 00m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/4/smb/ : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":4},{"id":"a389c8be9ab19e8cbb2c29d53f4b8d903e5a3760","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-10 19:12:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/4/check/dsvm-tempest-cisco-zonemanager-job/6726c66/ : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":4},{"id":"177a943679db6b36a6fe4ba772798d1110f98e4a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-10 19:32:50.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/4/check/tempest-dsvm-full-focal-py3/8b64eba/ : SUCCESS in 1h 20m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/4/check/devstack-plugin-ceph-dsvm-focal-py3/38445e1/ : FAILURE in 38m 59s","accounts_in_message":[],"_revision_number":4},{"id":"cf9555c13f7a4c4ab6a7424ff16280d418549c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 19:59:38.000000000","message":"Patch Set 4: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94f60019829e4285a54621b56b8d2db1 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709074e3204a4624854fdd6ae5d0a0c7 : FAILURE in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b51bfaffd77450d83932dc307975092 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0c7a6b7f9e4944b82ad5b6bea7ce02 : SUCCESS in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a82fdd5a2674e28a31cfe064b34203d : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62954c259886494797f07f1b2b40abd9 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/74e848c47ae142db920ca862117016e3 : SUCCESS in 1h 10m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ce2e9e0f7f04cd6abaa34c36cbc9aaf : SUCCESS in 1h 20m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e97b3de59c6a4b74b37364500b395bf2 : SUCCESS in 15m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9340fb744eae4bc6ac865a7be5462538 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/83972ab756114a98bc347c29a14be75b : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/25d4cf7808e441e68305f8f26f4824e8 : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0bd75f6acc834b3fbb0d41db3de7af8e : FAILURE in 31m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcbc33eb5d774e2d8e91859e0ea2ab56 : FAILURE in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b973cfef96c74d619d7187596acd263e : SUCCESS in 1h 30m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b478491fe734331a34566f99b771557 : SUCCESS in 1h 40m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5efa5e98af0472992ca7095006c2ad7 : FAILURE in 27m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd9ede1f77fe4b69b8f92e2ad7d11317 : SUCCESS in 1h 17m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/11e6fc2b06e24b58a2f8c03ecb11a0bf : SUCCESS in 1h 09m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98a4c4c1f04f4afcb182c0c7dd687baa : SUCCESS in 1h 15m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a876d4c3917c4533a8bc4b60561eda9d : SUCCESS in 1h 49m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a6c43d94fd04350a8a17a72c095da58 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6ba66f9db613d17cc23e776dd1b080e54231d4ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-06-10 23:32:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d0c1d36de5d9d0e57ff0be1cfeb6e0d86fdd6426","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-10 23:33:28.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/5/check-not-vote/ext-cinder-zuul/409b359 : FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"1e2c9cf4b5ce44a5509f78381e0b62a487575231","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 23:53:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbb6fe131cbd4912b9425a9d6d751f03 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84602636258e4b46ab95d2bcbf1300ad : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cf6165fa74a50fe6cf34cbc01270a2be14ad3878","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-11 00:05:14.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":5},{"id":"b717ab1321d9fa5ddc4a89d104299c451519c179","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-11 00:20:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/5/check/iscsi-driver-primera/f5522f4/f5522f4.html : FAILURE in 46m 28s","accounts_in_message":[],"_revision_number":5},{"id":"046e18a398d21f21d9d9e5fedbe8ac356b4486dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-11 00:21:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bdd022e/bdd022e.html : FAILURE in 47m 41s","accounts_in_message":[],"_revision_number":5},{"id":"a9aded0647c999cab9c152683a559a166c6ffc0f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-11 00:37:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/5/check/dsvm-tempest-cisco-zonemanager-job/e35f049/ : FAILURE in 58m 53s","accounts_in_message":[],"_revision_number":5},{"id":"c51761fb8eb2b39b7adbe192f864f95947f7648a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-11 00:39:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/5/iscsi/ : SUCCESS in 58m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/5/smb/ : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":5},{"id":"f57c8b495a6734f7c058f6180bd3569aef1b909d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-11 01:02:29.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/5/check/tempest-dsvm-full-focal-py3/87d8033/ : SUCCESS in 1h 26m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/5/check/devstack-plugin-ceph-dsvm-focal-py3/8f4e825/ : FAILURE in 41m 49s","accounts_in_message":[],"_revision_number":5},{"id":"cef2a9ce7b0628a6abdff65ce9106d95bad95bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 01:32:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11e2c72de4c04e1f915ef71506c66246 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ff7d40a2234ba89ef861121dac8f2e : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d1a9fec4ea462180415def6f35b753 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e730d6214d814e17ba9cf9b0e3604e49 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ddfdfc29ba64029bee561173a0a6773 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511e746c4d7b40debe1f295e8c1eb728 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/f9dccdb0c0c84240b92922c1b04bf653 : SUCCESS in 1h 12m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5407517bebd141248bb7a4e2d3014201 : SUCCESS in 1h 18m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/51ea37ad309c4f06b150baa21651efeb : SUCCESS in 12m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c05eecf39f0e4834bb4b61329cc98896 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d43bece2eb140838ccf246117a51f27 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65721a3b62b446428271e13ed797634e : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25d050431c0d4af08cf262b67ce3a38d : FAILURE in 35m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fa7f77eca2f4705b2a970b62e114ab6 : FAILURE in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f447c743703e4ecab1a00f5ce345ddfe : SUCCESS in 1h 30m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/552932d42f5b41208f644b3f96b3862d : SUCCESS in 1h 37m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2adeb9d20dad45a0a391e4d46e655131 : FAILURE in 29m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f4f22e175b04db8802db871bb1efb08 : SUCCESS in 1h 17m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dd5b8bf5aa624d2b8bef8ff56d923d7f : SUCCESS in 1h 16m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db57abfb7a9448bab2d1196e15c67447 : SUCCESS in 1h 17m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3fe96e5582a451582bc1f3f06c3e9d7 : SUCCESS in 1h 55m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fe6a749c5804b0d8a304b8f31a58678 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":5},{"id":"0c5022cbf3e4b9199996c1a164934b9a6335f879","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-11 02:10:58.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-4/ns5_nfs : SUCCESS in 4h 33m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-4/ns5_iscsi : SUCCESS in 4h 38m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"99bc60d8eeeebb5ab0c289cb196c250afc2b0a3b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-11 03:22:20.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/5/check/DellEMC_XtremIO_FC/e546b71/DellEMC_XtremIO_FC/271 : SUCCESS in 1h 16m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/5/check/DellEMC_XtremIO_iSCSI/0060362/DellEMC_XtremIO_iSCSI/274 : FAILURE in 1h 50m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"ad7904d245a84379ee508accdc11751f12b14ae1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-11 03:34:11.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/5/check/DellEMC_Unity_FC/25055ab/DellEMC_Unity_FC/281 : SUCCESS in 1h 33m 27s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/5/check/DellEMC_Unity_iSCSI/1d52337/DellEMC_Unity_iSCSI/265 : SUCCESS in 2h 02m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"282da97157ebbc2e5285e1fcb371b73c570c0895","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-11 04:09:24.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-5/ns5_nfs : SUCCESS in 4h 30m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-5/ns5_iscsi : SUCCESS in 4h 29m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"e2ca7e230922aa078f7352291f6461e27bdbcffc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-06-11 18:19:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bd574988a9559283a4af184a03e279e7816a2e25","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-11 18:20:40.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/6/check-not-vote/ext-cinder-zuul/2c09456 : FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"ef74059283cbf572be6ced41e1269961fcc983fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 18:51:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ecb329977c841948f5164bd4bb9284a : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa2ec895c8fb47588e78f61021b75c50 : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7c2384f684d19549cdb19bad53770e6e56f87adf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-11 19:21:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/6/check/iscsi-driver-primera/0146044/0146044.html : FAILURE in 1h 00m 33s","accounts_in_message":[],"_revision_number":6},{"id":"65708be896515944b3f586fe5657ee9c385a95c4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-11 19:23:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/6/iscsi/ : SUCCESS in 1h 03m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/6/smb/ : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":6},{"id":"b31c481af898aa7c5ecbbdd6bd1ee35544d82563","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-11 19:53:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/6/check/dsvm-tempest-cisco-zonemanager-job/3d17730/ : FAILURE in 59m 23s","accounts_in_message":[],"_revision_number":6},{"id":"04a76cddba2c3dfb9ceea5c644dc57f3e6e36064","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-11 20:28:02.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9448b727d7194ffc91bc259c1866ee78 : POST_FAILURE in 1h 39m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/131f3ee5b779494faf28daa5c15babd9 : POST_FAILURE in 43m 39s","accounts_in_message":[],"_revision_number":6},{"id":"11802071abe7e971bc53a726b47f2af7e7d69e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:28:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e6b7383411b4104bcf9d75f36fc1704 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec9545ada3d432db6b467ae23d60e7e : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89c03708d7124470a40b9580bd14a786 : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/746039775cde47f8b63103ab9fd2a077 : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddfbf4444f964c2da80ec424a9d4ddc4 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f27b683ede4618bc34c09fda43cbbe : SUCCESS in 13m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/8ecbbbc2856d4038a6fe47e090cdadc8 : SUCCESS in 1h 03m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e967506e624c444fae4106ccec919963 : SUCCESS in 1h 19m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/179cc31f72604c678eb435fbc4351ace : SUCCESS in 8m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/452392184e654bae834a9309ea2db6e0 : SUCCESS in 13m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f1c8bed8f0c4b83a9faf2d28a7a6f20 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b22b6d48040491bb925f547e6d9bcde : SUCCESS in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e1c4428ebcbd4f5999117b77c45b656d : SUCCESS in 11m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed3f7c2f5cd7462aa7715d132450c557 : FAILURE in 31m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d71e1b6193aa46c08a38fe999a7e4b4e : FAILURE in 8m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd081849d478467680c754c8b0c8b41e : SUCCESS in 1h 04m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/326532cb2d794ef985a48f86de4fa7ea : SUCCESS in 1h 42m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4967c37130ad4a9cb5f5a7f7d7df327d : FAILURE in 30m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/871682163b724846ba176c66daf31673 : SUCCESS in 1h 18m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/12443a36c08d41e0b126fedae553d419 : SUCCESS in 1h 37m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a720505cf1da403fa163363408b960ac : SUCCESS in 1h 15m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d986598012214f5db60916dbdc9e1750 : SUCCESS in 1h 46m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee40ebf95b54420583d2438813989aaa : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":6},{"id":"78d0d0c41519bb65faa819aa04ea49c30823c28c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-11 20:39:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cc0b4f3/cc0b4f3.html : FAILURE in 1h 59m 51s","accounts_in_message":[],"_revision_number":6},{"id":"f825bc95ab1a92dd3cdfdd46ab19437133f83dff","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-11 23:37:15.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/6/check/DellEMC_XtremIO_FC/20217ee/DellEMC_XtremIO_FC/276 : FAILURE in 1h 33m 38s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/6/check/DellEMC_XtremIO_iSCSI/dd493e3/DellEMC_XtremIO_iSCSI/279 : SUCCESS in 1h 45m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"afc43d676d80ad381f9e65def2f7828b2b0d8de7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-11 23:48:16.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/6/check/DellEMC_Unity_FC/0f2e275/DellEMC_Unity_FC/286 : SUCCESS in 1h 28m 29s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/6/check/DellEMC_Unity_iSCSI/4bd01f7/DellEMC_Unity_iSCSI/270 : SUCCESS in 2h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"7c27a78b357923acb3759ced51579512cf74fbd6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-11 23:55:36.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-6/ns5_nfs : SUCCESS in 4h 27m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-6/ns5_iscsi : SUCCESS in 4h 31m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"dc39583634cd691e6ec97d9b1d5745c51ef3d1bc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-12 08:14:14.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/60/786260/6/check/DellEMC_PowerMAX_FC/17acf2a/DellEMC_PowerMAX_FC/127 : SUCCESS in 4h 01m 41s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/60/786260/6/check/DellEMC_PowerMAX_iSCSI/4503094/DellEMC_PowerMAX_iSCSI/115 : SUCCESS in 4h 22m 37s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"4b4673a9e864a04c46d45fcbb2d0111f13143101","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-06-17 17:10:06.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a6fc5393ac7d9920375841d6afc15d54568aa956","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:29:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0edd5986063e46d79e2e90b6f86a0e43 : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a14afbed32a5497f8a3a8331cac1bd83 : SUCCESS in 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ac5d4a5a821e6d71dfa3cfc99bfb60cbd60b7521","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-17 18:12:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/6/check/dsvm-tempest-cisco-zonemanager-job/d61342c/ : FAILURE in 56m 40s","accounts_in_message":[],"_revision_number":6},{"id":"439a674d6e97cbe707562deac1865086fc5c9ece","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-17 18:14:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/6/iscsi/ : SUCCESS in 53m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/6/smb/ : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":6},{"id":"aad044d6e86f619c8cf7af904894e7291585be16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:06:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5280feeeb92479186f662c725587a13 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d99b63d4304709830c03ddc558f775 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/570188baa94d42409a7b0b276b120f66 : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69a443b4acf54e7a9d4c73921686554b : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed1131a1afe04e06ab3c8d7fba676cca : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b243cd0a346d4250a80b2174a3178193 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c0947d4e489e4fd1862d36b07c8d067e : SUCCESS in 1h 21m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf880476482f4e8d886cf97f27a9a699 : SUCCESS in 1h 08m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e0c352fe6bb415c876ebf0c6018d8de : SUCCESS in 6m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/99786dde0d8346099055c77fe3321f45 : SUCCESS in 15m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d143d6da2f24d76941d308494abb49e : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dea026235334956b9173cf56aa29611 : SUCCESS in 8m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9dd1f71578f04fbb92f0bb705863027c : SUCCESS in 7m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b25690ba29794b319562ec88cce7d8e3 : FAILURE in 36m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/46512764684749cda6b9dfcc09561319 : FAILURE in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/233ee07dee4a4b11880a5a3c4cc34389 : SUCCESS in 1h 19m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/39f1c85156284c32b76885bf1de0a4a7 : SUCCESS in 1h 39m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/254765b0293945c997455b8cfb29bbe1 : SUCCESS in 1h 24m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c5d3e7393b4e408ebba62744ec7aeef5 : SUCCESS in 1h 35m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e3e77e1046f40b3b20bb427cecec641 : SUCCESS in 1h 04m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/941c63e501734c3c80f2cef111cc89fa : SUCCESS in 1h 18m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87b33bf662d64ee197f68b42a01b01dc : SUCCESS in 1h 51m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c7d7ea1a6c1442bb93b6b4cd27a0fbc : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":6},{"id":"5f6baa07eb786b6597f5180f473f0ea593661750","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-17 19:20:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f9849d0/ : SUCCESS in 1h 15m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3f13419/ : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":6},{"id":"bd80b6026d879fb2cb6e301e758ebfcf956fdc0c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-17 20:06:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/786260/6/upstream-check/cinder-cDOT-iSCSI/69cd125/ : SUCCESS in 38m 40s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/786260/6/upstream-check/cinder-cDOT-NFS/e6ba47b/ : SUCCESS in 39m 39s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/786260/6/upstream-check/cinder-cDOT-FCP/ec564ff/ : SUCCESS in 57m 20s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"747f2c01973c521d152ab0147cd00a4a151f0903","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-17 20:14:10.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/786260/6/upstream-solidfire-check/cinder-solidfire-iSCSI/5892eaa/ : SUCCESS in 1h 05m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"7aadfa2ea18fdc592a49826e847479bbd92d212f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-17 20:36:57.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a1b63be987fa4ebfbb706d05af4fa635 : POST_FAILURE in 1h 44m 10s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7477a781a70f4184a2b7d0450f8898a9 : POST_FAILURE in 55m 11s","accounts_in_message":[],"_revision_number":6},{"id":"33cc2768e62c788b523e5bb66d31189a44b68de6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-17 20:50:17.000000000","message":"Patch Set 6: Code-Review+1\n\nLooks goot to me!","accounts_in_message":[],"_revision_number":6},{"id":"9fd30f9cc0afb994c0f04a2e6603028349ecbb2a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-18 17:46:47.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/6/check/DellEMC_XtremIO_FC/41cbaaa/DellEMC_XtremIO_FC/326 : FAILURE in 1h 25m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/6/check/DellEMC_XtremIO_iSCSI/a02f027/DellEMC_XtremIO_iSCSI/329 : SUCCESS in 2h 12m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"542bd3eee0e39408cb4e1e76590fcbd1ec205c6b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-20 23:40:28.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786260-6/Lenovo-iSCSI : SUCCESS in 45:58.65\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786260-6/Lenovo-FC : SUCCESS in 42:57.35","accounts_in_message":[],"_revision_number":6},{"id":"45da3070252e7f746cc603ced7897d0ffd420a1f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-20 23:40:32.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786260-6/HPMSA-iSCSI : FAILURE in 58:34.30\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786260-6/HPMSA-FC : SUCCESS in 30:58.89","accounts_in_message":[],"_revision_number":6},{"id":"a5c16613fa401df70485cef9dcb0dbe82aed7079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-06-21 03:45:11.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"4132b8607dfe6491f6940c15e8577be61def4adb","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-06-21 03:45:23.000000000","message":"Patch Set 7:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":7},{"id":"08e571630678aa1fb144e273d75fc1e5746a9077","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 04:05:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/357a9586e61e4459af8fd75243e1771b : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c6005a83c9c4c6c873ca0ebff764102 : SUCCESS in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f640c4b146888d7083d2d4180bb82197b701969f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-21 04:17:13.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/7/check-not-vote/ext-cinder-zuul/60e4051 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":7},{"id":"59de5bda41864d987acb87a850003d9d6445fdfe","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-21 04:54:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/7/iscsi/ : SUCCESS in 1h 08m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/7/smb/ : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":7},{"id":"843b9e5e2d23f699492010b3d55d0844ea7c6426","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-21 05:21:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/710fe85/ : SUCCESS in 1h 27m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a369f40/ : FAILURE in 1h 32m 19s","accounts_in_message":[],"_revision_number":7},{"id":"58997335777c83b5fae04f1cd6c5dc66855292c3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-21 05:45:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/7/check/dsvm-tempest-cisco-zonemanager-job/a8a5c58/ : FAILURE in 57m 04s","accounts_in_message":[],"_revision_number":7},{"id":"6389bd295fdbf46e31dd190e07f22aa38843a87f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-21 05:52:52.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/7/check/tempest-dsvm-full-focal-py3/7e0d38b/ : SUCCESS in 1h 47m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/7/check/devstack-plugin-ceph-dsvm-focal-py3/fb9bf3d/ : FAILURE in 1h 12m 17s","accounts_in_message":[],"_revision_number":7},{"id":"95ff771209ae099675972dfc9bc27d34af5afc97","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-21 05:55:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/7/check/iscsi-driver-primera/ae17b8a/ae17b8a.html : FAILURE in 1h 01m 20s","accounts_in_message":[],"_revision_number":7},{"id":"882fadd327fbf5393b9f5c9326052add6cd76a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 05:57:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79f6fb9c4cb84926b3bbe52686945d7f : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7372629805af402e85a53f502d42cc23 : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/177cd2ff8f184bee8e87e82381357180 : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/823074f08b41426ca75250a9754c1ff1 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc8c4bd632874f408ce91bb22a4de537 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfee7e8bff442559931811fe950854b : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/3cdb161930b24e2d8fee34619ef50933 : SUCCESS in 1h 00m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8783b723fd4246d9b8e7bc2b4daf2a4b : SUCCESS in 1h 07m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d4df09976344a32bc5a0cba70762322 : SUCCESS in 6m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a502c86eccbb47d48a283746b7ac22b2 : SUCCESS in 14m 13s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dcaea88c6f314600b50450fe915089e0 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/671954bda0624c8196bae678e49a2033 : SUCCESS in 7m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73fe2a06f4974de1a4a302f3aa7f306a : SUCCESS in 7m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1fd9bfe6debc4c55a1122196c02991ad : FAILURE in 28m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c30d606b7c514791bb1c0689ee027b8b : FAILURE in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/576487ff529947ed8c5a8cdb85606c48 : SUCCESS in 1h 31m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1510d9f995384f4ba87c97bb6cdb4a16 : SUCCESS in 1h 23m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bd60e4e7c0f4996add45a16991140df : FAILURE in 1h 15m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16d6e6ba0f354770a8566f947e4695cf : SUCCESS in 1h 13m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1fa88330ea164966ba4c3953d1e9345f : SUCCESS in 1h 08m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd745e2bd66d41fcba9bcbc054388078 : SUCCESS in 1h 23m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ca092e386ff4d0db59e08c371d7edc9 : SUCCESS in 2h 08m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95fd9e6c9eff43828eb30dec56f8ec3c : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":7},{"id":"4a5e90dffc2dfdb131cc7ac636933bb6a51bb1b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-21 08:01:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ba439dd/ba439dd.html : FAILURE in 1h 53m 31s","accounts_in_message":[],"_revision_number":7},{"id":"ad27bbadfc25d2ff33cbdb46ddbf92d67883396f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-21 09:42:06.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/7/check/DellEMC_XtremIO_FC/297c5f6/DellEMC_XtremIO_FC/350 : SUCCESS in 1h 22m 24s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/7/check/DellEMC_XtremIO_iSCSI/1b3c7fd/DellEMC_XtremIO_iSCSI/351 : SUCCESS in 2h 05m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"0a87ad4f1a5b4ddddea9e916ac2b343fc47f44dd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-21 09:51:37.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-7/ns5_nfs : SUCCESS in 4h 33m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-7/ns5_iscsi : SUCCESS in 4h 38m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"e754712050d5ef2505de80c0fece15a0212e637c","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-21 11:12:09.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786260-6/PVME-iSCSI : SUCCESS in 53:21.33\n* PVME-FC http://os-logs.tristero.net/dell-ci/786260-6/PVME-FC : SUCCESS in 53:02.79","accounts_in_message":[],"_revision_number":6},{"id":"f028dca63c5fbf1dd5f707916579097f6f12f56e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 10:41:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/7/20210622164324/NEC-FC : SUCCESS in 2h 26m 14s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/7/20210622164324/NEC-ISCSI : SUCCESS in 2h 58m 31s","accounts_in_message":[],"_revision_number":7},{"id":"9794e3d43f92ecc4e3a46376898f6fb1e5283a3f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-29 13:00:51.000000000","message":"Patch Set 7:\n\nIn addition to the unit tests, should we be able to observe any effect of this patch on the higher level testing, like in the cinder-plugin-ceph-tempest job? I think that job works even without this patch, so maybe some test there is missing? (but if it is, it can be addressed outside this patch, I\u0027m just curious).","accounts_in_message":[],"_revision_number":7},{"id":"edfa9559754c5d7022b4189ae3ad4bcf3c887fa8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-29 13:01:14.000000000","message":"Patch Set 7:\n\nrecheck\n\nsome fixes have been merged to make the iscsi job more robust","accounts_in_message":[],"_revision_number":7},{"id":"56d92112c4e5e08e4c5ba06390bf0b9385471b91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 13:29:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a626254c9404085b36525a9ef0f519f : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55fbb9988a6f444db73379152974cca8 : SUCCESS in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7182c90abebf21f24b786dd2030e5702b0a4190e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-29 13:56:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/7/iscsi/ : SUCCESS in 54m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/7/smb/ : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":7},{"id":"bb2fe90f0ebc75065d9301e5c3c5ebf027f97123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 15:03:14.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f8cf287bd341449d008bcbf8055fe9 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ff030d64a245168408841cfde233c5 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c57251518f4489ea7088fdb9f7fc044 : SUCCESS in 16m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/773ad325312443719b8d5d50fd221452 : SUCCESS in 18m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc90d112f6b7490b9c6bb8d62da103ff : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897ccdda27fa4a7a9a119c4071a57ad5 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a299704dcbf146319524b2850243aeb8 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94117f229c7a4d70a426950faecdcf58 : SUCCESS in 1h 06m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59308035a9b94eb6a4d91b2ac2b591b1 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd45d4b4016a420a8db2962c51cd6d53 : SUCCESS in 15m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/784df9c9b80f489a90e577c0948665f3 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3561f99ca7c44064b0428bde9f0fa507 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/725f356febc34d6b8cad8f816c37b18b : SUCCESS in 9m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b8d1a794e864ddbafe0ecd8f7007282 : FAILURE in 36m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47c2caf4b29c428285479a64ed20bbf3 : FAILURE in 8m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35442aef29114ae7864f6271885aa520 : SUCCESS in 1h 19m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae99fce75045433db9a7819c12e5f6c4 : SUCCESS in 1h 36m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d49997ce9f44c2686f70155ec06cc8d : SUCCESS in 1h 32m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f120ba95ed1949b88299ea090c5fe617 : SUCCESS in 1h 34m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1518f5c3097248d69d43e89ff27d42ac : FAILURE in 1h 28m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/986221b02d0f49618a8f965a67d4d4da : SUCCESS in 1h 17m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a56a587c0f5641bf9b61afb52cd34b90 : FAILURE in 2h 00m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cdbea474e844839aea03670420f040c : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":7},{"id":"d2294bcf63b7c4d8d902b5b00caa067e20baf510","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-29 15:15:40.000000000","message":"Patch Set 7:\n\nrecheck tempest.scenario.test_network_basic_ops.TestNetworkBasicOps.test_subnet_details\n\n\nTraceback (most recent call last):\n  File \"/opt/stack/tempest/tempest/common/utils/__init__.py\", line 70, in wrapper\n    return f(*func_args, **func_kwargs)\n  File \"/opt/stack/tempest/tempest/scenario/test_network_basic_ops.py\", line 636, in test_subnet_details\n    self.assertEqual(set(dns_servers), set(servers),\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/testtools/testcase.py\", line 415, in assertEqual\n    self.assertThat(observed, matcher, message)\n  File \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/testtools/testcase.py\", line 502, in assertThat\n    raise mismatch_error\ntesttools.matchers._impl.MismatchError: {\u00271.2.3.4\u0027} !\u003d set(): Looking for servers: [\u00271.2.3.4\u0027]. Retrieved DNS nameservers: [] From host: 172.24.5.89.","accounts_in_message":[],"_revision_number":7},{"id":"db0375fc70ba2711960653982d234a88d398c7ba","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-29 15:20:54.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/7/check/tempest-dsvm-full-focal-py3/2584e9d/ : SUCCESS in 1h 44m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/7/check/devstack-plugin-ceph-dsvm-focal-py3/cbe8102/ : FAILURE in 31m 52s","accounts_in_message":[],"_revision_number":7},{"id":"b63c6534473fcb729f3ddb54152761ed6f50c802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 15:40:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe8efb11af804c289a975bf2863bbe73 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33321776f3484547aa5f9b4d16c7c6c1 : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"562c8880c120e37d6b989a826df9804a669fd696","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-29 16:52:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/7/iscsi/ : SUCCESS in 1h 04m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/7/smb/ : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":7},{"id":"ca6b0232e81cdc360c8c43b77a35607b19ead22a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-29 17:52:36.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/7/check-not-vote/ext-cinder-zuul/4d50546 : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":7},{"id":"2b84a669c84e88689b34ef5db1f986243637566e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 17:54:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b38e886c2c24c31abc5c4277446f0ee : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41247bef9acb42728e25aab370535a71 : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a681387c3db48ef9bd0431ba48e49d8 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8920a7ae6744524925ed4e587a72907 : SUCCESS in 12m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3866c828b9c94a0785dfc44a9557b645 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf27ad947dce48e0b99d4ee961104bdb : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/00adb0a14c8449ecbbf3a2af19f7fd82 : SUCCESS in 1h 15m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc34b9c4104747278e397a015426155d : SUCCESS in 1h 06m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b133a1d20a974f6eade048060c974c6f : SUCCESS in 8m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e4b014decd75479187f212638cb301f1 : SUCCESS in 18m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ba5076a0e4647b5bcd75d97c0b86550 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f23f6422be6d487487c7e11f7717cc31 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/392f7c07c513475da77e3b01a0c0da7c : SUCCESS in 9m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bff6d094f275457185aa03a5daa65510 : FAILURE in 39m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5994cb228f0149fabfafed64637b1c71 : FAILURE in 8m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76025e78565f4164a356200e07b33ce7 : SUCCESS in 1h 24m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b2d513e8f0db4c73b12ed89ffe6e3c80 : SUCCESS in 1h 42m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/865c73f5008e4f30830eafb0e2a831db : SUCCESS in 1h 24m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb100b1394eb412faf56e887a7b64b56 : SUCCESS in 1h 15m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71e6e6143bd6469ab46149ff00c2206f : SUCCESS in 1h 28m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c810c82a831416cb0cad0f44d2c3c53 : SUCCESS in 1h 15m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/200bfa924b1541fa97b6bf58651153ac : SUCCESS in 2h 31m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bdd28b1b7cc4440bc355a09251219b8 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":7},{"id":"37c26e6723ad446a337562dd2f3294728b8f15e9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-29 18:51:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/786260/7/upstream-check/cinder-cDOT-iSCSI/54c94a3/ : FAILURE in 40m 04s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/786260/7/upstream-check/cinder-cDOT-NFS/a019984/ : SUCCESS in 32m 05s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/786260/7/upstream-check/cinder-cDOT-FCP/4e72f3c/ : FAILURE in 8m 14s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"ae26ce839097bdd6bd695098f0bd4367941fb65c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-29 18:51:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/786260/7/upstream-solidfire-check/cinder-solidfire-iSCSI/ca3ca31/ : FAILURE in 41m 26s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"0c3d5cf25dd91632ac3645efc63cfc5349f43e95","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-30 00:18:26.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/60/786260/7/check/DellEMC_PowerFlex/acbbbd7/DellEMC_PowerFlex/251 : SUCCESS in 1h 35m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"51a0f5381089d4892890253de8d2e152a841fdc7","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-06-30 01:37:34.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786260-7/Seagate-iSCSI : SUCCESS in 39:01.94\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786260-7/Seagate-FC : SUCCESS in 36:23.88","accounts_in_message":[],"_revision_number":7},{"id":"19659844ae5b3512b026339fb12431c52b3d5e27","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-30 03:20:09.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786260-7/Lenovo-iSCSI : SUCCESS in 45:57.95\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786260-7/Lenovo-FC : SUCCESS in 43:36.21","accounts_in_message":[],"_revision_number":7},{"id":"613406e764efd05d1eb6e0673b2685fb90664013","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-30 03:23:59.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786260-7/HPMSA-iSCSI : FAILURE in 1:00:12\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786260-7/HPMSA-FC : SUCCESS in 33:36.98","accounts_in_message":[],"_revision_number":7},{"id":"fa1d75cc685ed2177ac3cd81ab0e1cda936013e5","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-30 05:30:28.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786260-7/PVME-iSCSI : SUCCESS in 55:31.96\n* PVME-FC http://os-logs.tristero.net/dell-ci/786260-7/PVME-FC : SUCCESS in 51:39.19","accounts_in_message":[],"_revision_number":7},{"id":"1c468867fe40a18ddcccc32585ab483ad9293168","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-30 07:10:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"ce0fe8d23f844cb9c762b890ace1bfbe03e49289","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-06-30 10:35:59.000000000","message":"Patch Set 7: Code-Review+1\n\nLooks good to me","accounts_in_message":[],"_revision_number":7},{"id":"9c34d1d03e1803264de661c757ae3bc6fab8ca4b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-30 14:03:03.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/60/786260/7/check/DellEMC_VNX_iSCSI/c8d3dce/DellEMC_VNX_iSCSI/212 : SUCCESS in 3h 47m 07s\n- DellEMC_VNX_FC http://publiclogs.emc.com/60/786260/7/check/DellEMC_VNX_FC/d908e05/DellEMC_VNX_FC/230 : SUCCESS in 1h 42m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"b7afd38f24027f7f05c92c48124034537ea3f7b7","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2021-07-01 13:58:44.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)\n\nThanks for the patch, I was also hitting this issue.\nI\u0027ve founded some nits in unit test. IMO it\u0027s better to fix them. 🙏","accounts_in_message":[],"_revision_number":7},{"id":"0ea31822fb9bd6059fc99fa6bb4eafb56cad534d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-22 14:03:40.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nA few nits in the commit message, otherwise this fix makes sense to me.  Please either accept the unit test improvement suggestions or say something about why they\u0027re not needed so we can move this along.  It came up on the ML today [0], so people are beginning to hit this issue out in the wild.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2021-July/023784.html","accounts_in_message":[],"_revision_number":7},{"id":"490196502f276a96d46aeb94e2d6ae48170a293f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 14:03:02.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b6e2541dc3ff05f02228fb87deb22382a9cf98de","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 14:07:55.000000000","message":"Patch Set 7:\n\nI\u0027m working on it now.  I\u0027ll update the commit message to be more\naccurate as well, it\u0027s actually the stripe unit that conflicts - but\nwhen left unset (as it is in openstack) the value is determined by the\nvalue or order, so it looks like like the order is the issue, but really\nit\u0027s the stripe unit calculated from the order that causes the trouble.\nI\u0027ll include this information in the commit message.\n\n\u003e Patch Set 7:\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e A few nits in the commit message, otherwise this fix makes sense to me.  Please either accept the unit test improvement suggestions or say something about why they\u0027re not needed so we can move this along.  It came up on the ML today [0], so people are beginning to hit this issue out in the wild.\n\u003e \n\u003e [0] http://lists.openstack.org/pipermail/openstack-discuss/2021-July/023784.html","accounts_in_message":[],"_revision_number":7},{"id":"e6a90211a9ac1de5d49132851ec0a730b8a205c1","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2021-07-25 14:16:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a5ba30ded35d52cf040782a654aa7ea7026bbadb","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 15:13:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eeab907c3992f4c6234a6bf3ff3f519570a9b90a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 21:28:36.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"51d66712da8704e773d7416c5c033ca8e4429728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 21:28:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d21808b48917ab5089bc771f7a1b502688cd615e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 21:36:44.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2b0a7eb5a280b41795fa2cc68d689a2797629dbb","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2021-07-25 21:40:49.000000000","message":"Patch Set 9:\n\n(1 comment)\n\n1 minor nit.","accounts_in_message":[],"_revision_number":9},{"id":"f8f7d85c419416655c813bab6f49b825835d3686","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 21:58:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1263ee611f64cbcb27bb21d86883c30 : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0518fb904a144b7db36a708a7b4d8fe5 : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"80357a77ee4b6bb6b12e5a646f7259ed2e8407ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-25 22:02:31.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"ccd1be7646da8fcda57a8938ec3871dcaa418e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-25 22:05:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"089facd6220e05f73db13a34981d16e151c78eeb","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2021-07-25 22:08:15.000000000","message":"Patch Set 10: Code-Review+1\n\nThanks! LGTM","accounts_in_message":[],"_revision_number":10},{"id":"6de52883f319555f0fff2dc77f72eddb679e9a68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 22:24:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/231fc16ec8b3452cadef5623b04dae3b : FAILURE in 7m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55ffd5de39384957afb53a9bb5857153 : FAILURE in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fadd1d396b527039fd2572208bb25178d6d8a5a5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-25 22:31:27.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"36afb2edc91350f0c7dd5161ee4e9feb85508810","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-25 22:44:12.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/10/check-not-vote/ext-cinder-zuul/c967d74 : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":10},{"id":"3a73f9436e7ad5104e35d7bbef7999883c8092ba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-25 22:57:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/10/iscsi/ : SUCCESS in 51m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/10/smb/ : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":10},{"id":"0a6edc250f8b729107f31c7a07ebb719b4d8b9d5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-25 22:59:41.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/786260/10/ : FAILURE in 49m 02s","accounts_in_message":[],"_revision_number":10},{"id":"515ca0b98a1243b34fb26e6d74e278d3d5832acd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-25 23:10:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/10/check/iscsi-driver-primera/4363502/4363502.html : FAILURE in 1h 02m 59s","accounts_in_message":[],"_revision_number":10},{"id":"2c666d4ff92dc5975531a6d78263fd75c323d44b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-25 23:11:51.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/10/check/dsvm-tempest-cisco-zonemanager-job/8e26be1/ : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":10},{"id":"c2275b15dc7d0b02c33f97ed4f5c7ce6eda0572f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-25 23:22:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a14b8f7/ : SUCCESS in 1h 16m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7951592/ : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":10},{"id":"ab84d291c0eb465a2290dd884dce92425ed646c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-25 23:28:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c4aed61/c4aed61.html : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":10},{"id":"23f35e67d95f2993e3079ced759b188285a78fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 00:04:08.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02cf37a60da44339399273e82236b58 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/366d7bc2b8894767a3215b3e4f38375f : FAILURE in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b50951761e124e98b1b7306f010183aa : FAILURE in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adb61a07751b445c96003eded8d103af : FAILURE in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b4642774ce4b5d8b4313ade4ff5f94 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9ff94c36aeae408385af1d9db5b1b3f6 : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/624676727dd844a3a95299e405c1a44e : SUCCESS in 55m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e1818637d4c485c89c93710d686917d : SUCCESS in 6m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7b9375129a5f46a5bf24793b4227f62a : FAILURE in 5m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/821754b445d640bbb52244bed3971ea4 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37c5175fda9044168cd39d9a55bee4f1 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/806be54434c9457d91d4ca36d55d0ffc : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bee1b50e78844a5db537b82a162660c8 : SUCCESS in 29m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08772d11e8164a5a991bb54595128f90 : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ff845a2a30f4252b58986b13ae5d46a : FAILURE in 1h 08m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c074bba3fcc643d9ae5b4b42321b6980 : SUCCESS in 1h 38m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0ac6fac45f9457cab7cbede49b5fe1f : SUCCESS in 1h 50m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df57d37d533e497a979aa022b12cf667 : SUCCESS in 1h 02m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6ebfdd0076f41fc88c9b1f392d10148 : SUCCESS in 1h 15m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7a0d7e5fedcd49699cded77aa79c7c80 : SUCCESS in 1h 20m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e50f2cf7cfc4270840b51a7eea7ee7c : SUCCESS in 1h 46m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5006787caaeb4294ad5dd0b4e415446f : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":10},{"id":"304f8681f746ca0b5a2a9f8dccdda0707f1c82d2","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-26 00:12:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/8/20210726062853/NEC-FC : SUCCESS in 2h 37m 58s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/8/20210726062853/NEC-ISCSI : SUCCESS in 2h 43m 31s","accounts_in_message":[],"_revision_number":8},{"id":"916a808a389b3af0e703e259e2cc02840130fa75","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-26 01:37:56.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-8/ns5_nfs : SUCCESS in 4h 4m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-8/ns5_iscsi : SUCCESS in 4h 8m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"a4d4c7bc98b50ac26a586331e20b1ed16103b81b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-26 02:26:15.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-9/ns5_nfs : SUCCESS in 4h 6m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-9/ns5_iscsi : SUCCESS in 4h 11m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"e7d62e2fb2eff340ddb22b341dac8939f9d5d96b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-26 02:59:36.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/10/20210726091233/NEC-FC : SUCCESS in 2h 27m 35s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/10/20210726091233/NEC-ISCSI : SUCCESS in 2h 47m 00s","accounts_in_message":[],"_revision_number":10},{"id":"3050f593fe64ee3c2ee368f56edbb46f6f83b929","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-26 03:08:46.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-10/ns5_nfs : SUCCESS in 4h 6m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-10/ns5_iscsi : SUCCESS in 4h 9m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"2f4c7739826285abde08209aaf983fca9411664c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-26 15:23:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b651326d6ab3dfafb66788bfe35da78f7aca7edb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:42:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69c59b2c5294404dbd8ed8c4596c9c7c : FAILURE in 10m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/645517030dda4638b20df953fec75558 : FAILURE in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"baaaf9994edd0cbba1d0974eaadcebf3de1618df","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-26 16:11:43.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/786260/11/ : FAILURE in 46m 53s","accounts_in_message":[],"_revision_number":11},{"id":"5d9ff4b07b54630fabc409ed2184157ac8be0cb0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-26 16:23:06.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/11/iscsi/ : SUCCESS in 56m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/11/smb/ : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":11},{"id":"d4ea0848a3ff26284f8e5bd9fef6fae514514c22","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-26 16:25:57.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/11/check/iscsi-driver-primera/5ac952a/5ac952a.html : FAILURE in 1h 01m 22s","accounts_in_message":[],"_revision_number":11},{"id":"5d2974a5d33ed2eeb9ad6c942ba98bb7dd2f2cd9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-26 17:23:44.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d713f91/ : SUCCESS in 1h 12m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/11/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2e6e1a5/ : SUCCESS in 1h 59m 23s","accounts_in_message":[],"_revision_number":11},{"id":"7f194a820f83273be7f21eeb0927dea7c9607acb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-26 17:32:14.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/11/check/dsvm-tempest-cisco-zonemanager-job/4a16b47/ : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":11},{"id":"77814012174150b8d563488663e861024e5b260d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 17:43:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9555cdfccec945a9b1cc2de40b97f215 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcbc893ebdfc405792ed4f9ec7e7819b : FAILURE in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/befef4b184d54bacafaab743ff62b7ac : FAILURE in 11m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57dafc27095c40ea8842674b655cedb0 : FAILURE in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5f648d53d44e9d932a2b08d9eedaaf : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/aa351fd5c90d473cad3efff721bde88e : SUCCESS in 56m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35408f9dccaa4d2288c0ce5c9a8a01c9 : SUCCESS in 1h 08m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d61cb89c45a6437faf4c2ebdff164121 : SUCCESS in 9m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c2dad7f76d54f3c99077adeae5d731a : FAILURE in 11m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aeae0a8bc00d4555bfc46b9b2a7a10c1 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c81102d2b7a4fd3b603587ffa1df64f : SUCCESS in 7m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/802a0bf8568640e69b06032c20560fc4 : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b451400fbfcf4ebe92fd97d5011a2be5 : SUCCESS in 34m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/df0cedcf7dff45d2a1511422b439e737 : FAILURE in 8m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eafcfbac0be54f90924e3a429360c806 : FAILURE in 1h 12m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b859f88d19a46539ef297e4c50e652a : SUCCESS in 1h 23m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e43f9afab1a24cda821f74d6929f17ed : SUCCESS in 1h 13m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b4e2d643458c4a4bbb08d3d6ffb545f6 : SUCCESS in 1h 03m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e8bc3b43c92c4d4193beeb04ccccaed4 : SUCCESS in 1h 05m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/055e4bd7698c47ffbd7e28aa3085f735 : SUCCESS in 1h 20m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/266de665fff84a8c97b39ec19061f550 : FAILURE in 2h 15m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62a3b00cfca84cb19db4a0d63a4a03fc : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":11},{"id":"6e6a9e4fe1091850a37cf9d4eec1f0553106368d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-26 18:01:11.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4479b36/4479b36.html : SUCCESS in 1h 17m 56s","accounts_in_message":[],"_revision_number":11},{"id":"d03d497415489f588f090eadd511dc2bd7b14c2f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-26 18:41:39.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/11/check/DellEMC_Unity_FC/0be6cd2/DellEMC_Unity_FC/501 : SUCCESS in 1h 07m 16s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/11/check/DellEMC_Unity_iSCSI/695ca33/DellEMC_Unity_iSCSI/472 : SUCCESS in 1h 35m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"993047369e9461e85379f28b195271c2a4ffe448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-26 18:58:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0beba281dd1ca4f00335d98587f777ee4f46cca3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 19:20:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ce64bff80f9401fb07c15d7ff700cfa : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a5bc1bcda5f40d88dd02294a21d4127 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"eb5e7e3f6f7d3d27a555293e02793237e856dac2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-26 19:52:15.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/12/iscsi/ : SUCCESS in 52m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/12/smb/ : SUCCESS in 47m 50s","accounts_in_message":[],"_revision_number":12},{"id":"acb6a76e5e626042cce2f2b61649209ed99c2e4d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-26 19:54:24.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/786260/12/ : FAILURE in 51m 04s","accounts_in_message":[],"_revision_number":12},{"id":"14f4e45a2111f0d2bc63fbd2736074c13610790d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-26 20:02:10.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/12/check/iscsi-driver-primera/4d61ead/4d61ead.html : FAILURE in 1h 02m 00s","accounts_in_message":[],"_revision_number":12},{"id":"4472f7b8a35f4e7c99ab069ecff532e8f6ce9739","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-26 20:08:00.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/12/check/dsvm-tempest-cisco-zonemanager-job/9e53896/ : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":12},{"id":"08d85dd833651fe0ae1c348318a5fa7110ced313","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-26 20:19:52.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/de3e722/de3e722.html : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":12},{"id":"044c6b517214dedea6e95e60fe78bae4fc33a9a1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-26 20:33:09.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a51d616/ : SUCCESS in 1h 16m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4816a16/ : SUCCESS in 1h 33m 00s","accounts_in_message":[],"_revision_number":12},{"id":"5d893239bd27937037d73f91b2e2a72675823e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 20:53:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7c8ac583164d9c95c21ef118ec708d : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f09d132848db47298a913443c1fe8f98 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2569f28bde2454e9151d10ca0deca16 : SUCCESS in 16m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7133f1178dfb4fd0b30c9d245f731a0e : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b484615e9504fe1aa21ed0d0e4bb3c3 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b079040ea4d0412297d58300ead4c34a : SUCCESS in 54m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b44dd12f0cab400c9523a0a9f6be0a0a : SUCCESS in 57m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c5dc93d51aa4b5094b4639b066f4bab : SUCCESS in 6m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d84f74615734cb995684e4b4f7d50cb : SUCCESS in 10m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c967b65c5ea472f88296d116666d915 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc2d64a372914161954b7726fbd96ce9 : SUCCESS in 6m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7bce19153fc946ce8b8341a7f15c070f : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7ecdc2def9b549bb88350ad2c7e9c75e : SUCCESS in 35m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7255561db884f23ba1b6db6d213eead : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea7c5effef8d41d799a3fcd5d1f895c5 : FAILURE in 1h 08m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b57f988ab26543c793a0ddd472f7d940 : SUCCESS in 1h 38m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a1d1129ead8456098266ff80704fdf5 : SUCCESS in 1h 23m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7c1a27ab6c34bb18534d90ca118b447 : SUCCESS in 1h 08m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/faf9aac2d7534b2b949915919903e3d7 : SUCCESS in 1h 03m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9173e0131c124430a398c0fbdc920b29 : SUCCESS in 1h 09m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93aed0a9a25e4905a294f71f2f30c0f1 : SUCCESS in 1h 52m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df918331dcd2499c926ada6043e829ab : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":12},{"id":"cd84e49d0db8f04b186837518de0b9003048f01a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-26 21:40:14.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-11/ns5_nfs : SUCCESS in 4h 17m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-11/ns5_iscsi : SUCCESS in 4h 22m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"e8a68d5981347b99911f15d55502513e54985da9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-26 21:50:14.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/12/check-not-vote/ext-cinder-zuul/8eee392 : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":12},{"id":"4a18b6c2026596803af1f5ae9c521ec69024828b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-26 21:51:11.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/12/check/DellEMC_Unity_FC/253edfc/DellEMC_Unity_FC/503 : FAILURE in 1h 09m 03s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/12/check/DellEMC_Unity_iSCSI/e91ab1d/DellEMC_Unity_iSCSI/474 : SUCCESS in 1h 36m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"6f2343a3f54acd655e1a99ee0d6dfccd6d433f5b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-26 22:46:49.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/12/20210727050413/NEC-FC : SUCCESS in 2h 28m 06s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/12/20210727050413/NEC-ISCSI : SUCCESS in 2h 42m 32s","accounts_in_message":[],"_revision_number":12},{"id":"1051c9ee55002be8f6bd03dd52f5427198d02d10","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-26 23:16:08.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-12/ns5_nfs : SUCCESS in 4h 13m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-12/ns5_iscsi : SUCCESS in 4h 16m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"39e7f34a4685a459ba47b189d02a5db33e268ebb","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-27 02:47:00.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/60/786260/12/check/DellEMC_PowerMAX_FC/69376eb/DellEMC_PowerMAX_FC/277 : SUCCESS in 3h 31m 27s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/60/786260/12/check/DellEMC_PowerMAX_iSCSI/5d4598e/DellEMC_PowerMAX_iSCSI/270 : FAILURE in 2h 56m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":12},{"id":"e2119f1d813f1c983793718f4d25b7ffbbae8057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-27 17:17:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3902d03e6bad168a8e7c0c5828a27546931313c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 17:45:52.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3601582f4ab42d7a516a26db3440a18 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40f1e0bd60ee4876bad07bea02efd03e : SUCCESS in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fa7b191c1259dc4c63c4be0f14b52948442945ac","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-27 18:12:56.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/786260/13/ : FAILURE in 50m 42s","accounts_in_message":[],"_revision_number":13},{"id":"5497c170054673cb184c0820ee1d54fa768f65ca","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-27 18:13:45.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/13/iscsi/ : SUCCESS in 53m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/13/smb/ : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":13},{"id":"67660af582edfcd83ea3a3f3afea549e01543f70","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-27 18:24:12.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/13/check/dsvm-tempest-cisco-zonemanager-job/09cf302/ : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":13},{"id":"221a31cc40373cb44f27c8508beae20a8e6ecc77","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-27 18:24:32.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/13/check/iscsi-driver-primera/93f6227/93f6227.html : FAILURE in 1h 03m 29s","accounts_in_message":[],"_revision_number":13},{"id":"dd8e76d4dff0a8fadb9476a1eafbc78cb351c992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:35:16.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb42795e6f14a2aabf2237fc087b632 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da7207ef941742a9bb9b771d57a56fcf : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0144798f17244b5287741645da8fb403 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d4143631c4c4c83a94e2c4cbc16f3ff : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9497800811374f1faee1d413b19d2348 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/86d17113976f44d2b3ecb2d20497f998 : SUCCESS in 59m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b895ef68bbd544759615ed10a3272a42 : SUCCESS in 1h 08m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c907f591d2e49d6a4b7a27d57b93e65 : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a29444385934eed9913888a9cd9dab2 : SUCCESS in 12m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3acf45fd44154c3d82f3adc4646aa036 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa03bad3127a4e25825df36409751f4c : SUCCESS in 6m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ddae0e4147e4d8cbe070d95ca0f5674 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ab8ff47f992241d7a3119e38980caabf : SUCCESS in 34m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7de30b0870a448eaad130a81d2e64b6 : FAILURE in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44348a1851934ac68864ed8baa6a9124 : POST_FAILURE in 1h 19m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/984c0ee93d744fd3aa0f7bc113906fbc : SUCCESS in 1h 31m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1dc9d5d78184299a14d91a1e754dde0 : SUCCESS in 1h 29m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3f548709b7374c5fa110f1d1e262606a : SUCCESS in 1h 11m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe23ec549de441bfa142f0d0ede15054 : SUCCESS in 1h 15m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d4a306b2fa214b08a63ad200b27cea1f : SUCCESS in 1h 21m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82f1631a23e44036a542482e065e2154 : SUCCESS in 2h 12m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/073015b2794b4ddc958840a31fa35413 : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":13},{"id":"ebe9dfff9b6208178c2fce2457d56360ed54e22f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-27 20:07:29.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/499dc87/499dc87.html : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":13},{"id":"490630175033b7b91e1291acca5420f028efa405","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-27 20:27:20.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9bc654f/ : TIMED_OUT in 3h 08m 06s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7671400/ : TIMED_OUT in 3h 07m 02s","accounts_in_message":[],"_revision_number":13},{"id":"6983d723b0111c923c2c55469222863f949b37d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-27 21:07:23.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/13/check/DellEMC_Unity_FC/2e98445/DellEMC_Unity_FC/511 : SUCCESS in 1h 12m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/13/check/DellEMC_Unity_iSCSI/ada03b7/DellEMC_Unity_iSCSI/482 : SUCCESS in 1h 44m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"89812053a2515ecc55e64a732d172b7908f30318","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-27 21:25:49.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-13/ns5_nfs : SUCCESS in 4h 3m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-13/ns5_iscsi : SUCCESS in 4h 7m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"f43e239e82a875425148e9bf9314c9211086886e","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-07-27 22:13:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/c06844d/index.html : SUCCESS in 1h 02m 59s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/be8e8e4/index.html : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":13},{"id":"187c69b77ebe20d5d9a34cfde5612261fc37da50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-27 23:34:44.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/786260/13/check/iscsi-driver-nimble/f576854/f576854.html : SUCCESS in 1h 58m 27s","accounts_in_message":[],"_revision_number":13},{"id":"7fc99810320e04d6b6ec82dfd344be2f3e5caa2f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-28 01:24:01.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/13/20210728074119/NEC-FC : SUCCESS in 2h 28m 34s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/13/20210728074119/NEC-ISCSI : SUCCESS in 2h 42m 37s","accounts_in_message":[],"_revision_number":13},{"id":"316ca89761e00a579ae7e2d4058ce8b3122eabe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-07-28 02:43:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2bb8032e69059992de1a6d728aa748cf04dd0db6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-28 02:44:38.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"420b4db159aa07103e791410716eb2c3f4dbc9cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 03:05:31.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4408656aca34d389994e47cacf95c02 : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebcdf009f6f54b68815fe1aec6555020 : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c86e8a740236c5cafc3b6d52193abe0010283f59","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-28 03:35:41.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/786260/14/ : FAILURE in 44m 56s","accounts_in_message":[],"_revision_number":14},{"id":"8693f896f7b7e8ddcbec1e9104bb711bc59b303a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-28 03:42:29.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/14/iscsi/ : SUCCESS in 58m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/14/smb/ : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":14},{"id":"9d94a7ee9acdca3bd8981da9befa0f5420e307e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 03:48:22.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/14/check/iscsi-driver-primera/419b899/419b899.html : FAILURE in 1h 03m 01s","accounts_in_message":[],"_revision_number":14},{"id":"90a2adbfd2d483a9e287d329211afdde0bd30fec","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-28 03:50:50.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/14/check/dsvm-tempest-cisco-zonemanager-job/ca162b5/ : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":14},{"id":"ef6071bd11869105276778df34f235fe9b9816e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 04:04:53.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c0f7b16/c0f7b16.html : SUCCESS in 1h 19m 34s","accounts_in_message":[],"_revision_number":14},{"id":"b40285358ed80ba7513f62922b2c7d0e20555b51","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-28 04:27:46.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/14/check-not-vote/ext-cinder-zuul/346740a : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":14},{"id":"a5652b82233294863eb4c717d3cc593da430be25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 04:36:07.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1439010f89a34c6eac34dba870268381 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b2b0cf6eafc4b3bbb3913437bd77373 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32f82362991f4f88a9f25b3083546275 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d95ebd3d3fad4dc891833a28e1f12b9c : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de4ab3bd816498cae510720a5aa5fcc : SUCCESS in 18m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/539ed46d3b944309b078f15ff785512b : SUCCESS in 54m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc31d6a7c14a40869c116ceb034c9748 : SUCCESS in 1h 08m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce28a2d07dbc48e9bdb72abdae9c9012 : SUCCESS in 6m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d7f97e0ab3949d0b934534b1ea12344 : SUCCESS in 11m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c4f596ac2574c34bbdf8c7cddfafb1a : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a39efb42e3184617b17d128806ef0eb5 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a2ac20225e64e7b9391fd835058013a : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28671d58197042f69886959a70af5f01 : SUCCESS in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3cafa92bda049228239743e70a49dc9 : FAILURE in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d512b4449e5b4a8b8786d52afc5f24f9 : SUCCESS in 1h 14m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce6b89910e154d679e355516c4e1ceee : SUCCESS in 1h 24m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f579de9dcb3341eb8ab3d31b2d39646d : SUCCESS in 1h 30m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b65719d401a4d488117d847fa9d0e4a : SUCCESS in 1h 18m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/645ec8ba777641ff8221a1880fb4f8d4 : SUCCESS in 58m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57049d03d2a7457294dc82580b41d237 : SUCCESS in 1h 16m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c9374174f24a4c8f9c4bffee6b9faff6 : SUCCESS in 1h 50m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e1cf773192547d4b0d590c68c17fda3 : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":14},{"id":"bcfff584f4aba9f10322341e3fb3a878606a9f81","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-28 04:45:12.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/14/check/DellEMC_Unity_FC/3befd83/DellEMC_Unity_FC/513 : SUCCESS in 1h 32m 00s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/14/check/DellEMC_Unity_iSCSI/ec77ea3/DellEMC_Unity_iSCSI/484 : FAILURE in 2h 01m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"e47b61af4ccb2151d359f49417cd4b3d8754356c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-28 05:01:26.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/786260/14/check/iscsi-driver-nimble/5260cec/5260cec.html : SUCCESS in 2h 16m 44s","accounts_in_message":[],"_revision_number":14},{"id":"ce1c8ecb26f84eb7b0afa1cabf928ace02ea3073","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-28 05:57:18.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/60/786260/14/check/DellEMC_SC_FC/03a512f/DellEMC_SC_FC/189 : SUCCESS in 44m 07s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/60/786260/14/check/DellEMC_SC_iSCSI/5fa01fb/DellEMC_SC_iSCSI/185 : SUCCESS in 1h 20m 52s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":14},{"id":"630140b3a555885ff84d48412579888f6399eabc","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-28 06:21:03.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/786260/14/upstream-solidfire-check/cinder-solidfire-iSCSI/d7c2fc7/ : FAILURE in 1h 42m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":14},{"id":"40fff9e56039c89758e9d0d5d2722f2bfcbd0379","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-28 06:22:55.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/786260/14/upstream-check/cinder-cDOT-iSCSI/10abf0c/ : FAILURE in 1h 45m 10s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/786260/14/upstream-check/cinder-cDOT-NFS/05c6f53/ : SUCCESS in 1h 41m 28s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/786260/14/upstream-check/cinder-cDOT-FCP/27a64ef/ : FAILURE in 6m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"23fcf8e2bbfeea0c4a27a9e46eca7ffe8c0dcc6b","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-28 06:28:24.000000000","message":"Patch Set 14:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786260-14/Seagate-iSCSI : SUCCESS in 42:21.34\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786260-14/Seagate-FC : SUCCESS in 40:18.95","accounts_in_message":[],"_revision_number":14},{"id":"e1a87591563cebc29eeb1a2f3026b5446c909749","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-28 06:36:53.000000000","message":"Patch Set 14:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786260-14/Lenovo-iSCSI : SUCCESS in 46:21.97\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786260-14/Lenovo-FC : SUCCESS in 44:50.03","accounts_in_message":[],"_revision_number":14},{"id":"a0e62b2b89a7151ad2772c56844104df72ccaf8c","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-28 06:39:12.000000000","message":"Patch Set 14:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786260-14/HPMSA-iSCSI : SUCCESS in 57:57.64\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786260-14/HPMSA-FC : SUCCESS in 35:17.76","accounts_in_message":[],"_revision_number":14},{"id":"a544f048dc584e69547cb77886a304d7c616ccb5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-28 06:55:16.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-14/ns5_nfs : SUCCESS in 4h 7m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-14/ns5_iscsi : SUCCESS in 4h 11m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"5c3de1293959473256354bd98f05518e292a2c9e","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-07-28 07:05:41.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/b530a10/index.html : SUCCESS in 1h 01m 21s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/4376482/index.html : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":14},{"id":"006d36df30ad4b2065860e8f5036b2f573f5bef1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-28 08:16:45.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/60/786260/14/check/DellEMC_VNX_iSCSI/8a2ac47/DellEMC_VNX_iSCSI/268 : SUCCESS in 3h 40m 21s\n- DellEMC_VNX_FC http://publiclogs.emc.com/60/786260/14/check/DellEMC_VNX_FC/c84f038/DellEMC_VNX_FC/286 : SUCCESS in 1h 33m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":14},{"id":"264c6b4504bb9d4038aeed463d38ed527a6b7aa8","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-28 08:52:59.000000000","message":"Patch Set 14:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786260-14/PVME-iSCSI : FAILURE in 3:01:56\n* PVME-FC http://os-logs.tristero.net/dell-ci/786260-14/PVME-FC : SUCCESS in 46:00.23","accounts_in_message":[],"_revision_number":14},{"id":"ea85a0abc1e468f2bc9866b5ed802d470964b6ba","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-28 08:53:43.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/14/20210728130208/NEC-FC : SUCCESS in 2h 30m 14s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/786260/14/20210728130208/NEC-ISCSI : SUCCESS in 4h 51m 31s","accounts_in_message":[],"_revision_number":14},{"id":"583d0d10c1d7607aeb6afd20f7f5438baacf6090","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-29 13:05:48.000000000","message":"Patch Set 14: Code-Review+2 Review-Priority+1\n\nThanks for the detailed commit message.  Code and tests LGTM.\n\nRaising the priority because this has come up on the ML and people are encountering this issue in real life:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-July/023784.html","accounts_in_message":[],"_revision_number":14},{"id":"60f8b1c400ad4829aa93192e54a4abfe470c35f0","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-02 05:33:02.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/60/786260/14/check/dsvm-tempest-TOYOU-ACS5000-driver/ad7900b : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":14},{"id":"8f5af8f3bd2f6b7c4b3e8850922c6b27ff671044","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-08-02 08:07:01.000000000","message":"Patch Set 14:\n\nUnfortunately there is a merge conflict - can you please solve it?","accounts_in_message":[],"_revision_number":14},{"id":"d25751e09272c4dd5f7b39670a8cba217c9485eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-08-02 15:13:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"954228d79240745e4897f6170335ee751d842ea4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 15:33:36.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58fbc58c90d34989a28aa0036193a998 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02d08a249a994d16a2f47657326dabde : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a5f7e4a83f7c7f3fec7cf6bf0d5069f55548ba7b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-02 16:07:20.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/15/iscsi/ : FAILURE in 36m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/15/smb/ : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":15},{"id":"e1d15983a2ecae844f159e6a1d507bf6b3ec93c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-02 16:21:00.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/786260/15/check/iscsi-driver-primera/52d125a/52d125a.html : FAILURE in 1h 05m 11s","accounts_in_message":[],"_revision_number":15},{"id":"ab94eb6be9e76f4f3cfaad45e3c915bad1cead76","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-02 16:21:52.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/15/check/dsvm-tempest-cisco-zonemanager-job/b6a97ac/ : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":15},{"id":"9f5d0460204f7159de602c62c99adaa32f4bf346","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-02 16:37:12.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/786260/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4f7bd06/4f7bd06.html : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":15},{"id":"d348d8bfe6600a3254a9af75517699ec5bf9e298","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-02 16:38:37.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7dbacc0/ : SUCCESS in 1h 23m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/689bba2/ : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":15},{"id":"1ed479f262639bd4e9ecb7dea05c65392287e3d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-02 16:56:28.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/786260/15/check/DellEMC_Unity_FC/bc373f6/DellEMC_Unity_FC/549 : SUCCESS in 1h 07m 54s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/786260/15/check/DellEMC_Unity_iSCSI/e779b00/DellEMC_Unity_iSCSI/520 : FAILURE in 1h 21m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"87a702b12138c33b9f7c2c17f2c749b5cdd0bae8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-02 17:13:52.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/786260/15/check/iscsi-driver-nimble/37d2b49/37d2b49.html : SUCCESS in 1h 58m 36s","accounts_in_message":[],"_revision_number":15},{"id":"02c3373b515adf3fc37e24063a3113a2e455e2cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-02 17:21:59.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cbac2aeceed042a0935e30ffc345903c : DISK_FULL in 2h 04m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/15/check/devstack-plugin-ceph-dsvm-focal-py3/d6983b3/ : FAILURE in 41m 32s","accounts_in_message":[],"_revision_number":15},{"id":"1fd04f83df993f922efe3715b6dadd0b027751ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 17:27:37.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275fa68ff3ed4e73aac60648b3229c3d : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e885759282284c3d9aa687952a7021dd : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/158ae974ebe74b8c8078c01937662278 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/719a98245af1416c895a0acf5acc94ab : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f002371f634537a7e41e05b7f36e42 : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/93348ea21c53442f8d4b88343d77918a : SUCCESS in 59m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1aa8f3930a74d81acc7080e2a16778c : SUCCESS in 1h 18m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cce74703cbfc4b97bd2a061f87095562 : SUCCESS in 6m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7614b206eef24d79bb266ea444e8c188 : SUCCESS in 11m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00d39b03804e47b2ae9e22a4b8be2eb2 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fd19301ec30438f9bbb7fd71f94eaa5 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5abdad9e7a1f434988e280359f267bc6 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0317970b433e4d1ab603693b67f573d6 : SUCCESS in 34m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e7d475e8c524dab95ea2fa8d59867b2 : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ef3cf3b7cea0449c8112e74834babf27 : POST_FAILURE in 18m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/af0ba9576a7c4ca1ac86eb40fea71cb4 : SUCCESS in 1h 28m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1be6d8e2f36a45eaa7b708e8fb10dc32 : SUCCESS in 1h 21m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f179ae92174a4f43bd4dc9edd66eece8 : SUCCESS in 1h 19m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/14b280d2b2134695b0704d0d5206f149 : SUCCESS in 1h 01m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34ee4327d3004cdea542e432ffced998 : SUCCESS in 1h 09m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df3e8773be524859801b175f0baf6cf8 : SUCCESS in 2h 02m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba177e842d534a37ab5ee378868e9e96 : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":15},{"id":"a9534cf9ae26013afedf7a50722115878ed21bf4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-02 19:25:12.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-15/ns5_nfs : SUCCESS in 4h 9m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-786260-15/ns5_iscsi : SUCCESS in 4h 10m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"6f6a68064d414ca90576add22ea33912a12145d5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-02 21:00:17.000000000","message":"Patch Set 15:\n\nrecheck cinder-plugin-ceph-tempest -- error occurred really early in devstack setup, hopefully just a random thing","accounts_in_message":[],"_revision_number":15},{"id":"63cf95a7635a52b1338ff0f710f138c6dcc5fe0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 21:24:38.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4505e2aff7a648d18f0bf7b48eb17019 : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bbb9e9ee6c242bc896371c4ed81c683 : SUCCESS in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4764197927648702fb45f7e1bc6b7e6219e09952","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-02 21:51:13.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/15/iscsi/ : SUCCESS in 50m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/15/smb/ : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":15},{"id":"4125f54360717e896dae982e0377c0e54d7dde2f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-02 22:54:28.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/15/check/tempest-dsvm-full-focal-py3/bb9f879/ : FAILURE in 1h 50m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/15/check/devstack-plugin-ceph-dsvm-focal-py3/bfbe174/ : FAILURE in 42m 49s","accounts_in_message":[],"_revision_number":15},{"id":"0d068918ad19c887e0f12398797510bb751d117d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 23:00:16.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1457d78bde4cb68ca5478cf45375a7 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d86ad5d69c08468ca72b39e5668cc7cf : SUCCESS in 14m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdab56903a284ee99aa88068ab0fe80b : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e5b5b8370474f51ae4827d13a1b5e0a : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a09e9886a94c2da51300a81b21883a : SUCCESS in 16m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/120ac59a1e6e4b00a5163977d636ed1c : FAILURE in 1h 03m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/add220e2795744049d6613723451855e : SUCCESS in 58m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ae9b78441694f74a3bd90c8e8709940 : SUCCESS in 6m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b5e7649361d94fdbbab1cf6edd5010d5 : SUCCESS in 10m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72a71c0da69d4387a9a1ac8850f201fe : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7b826f1a05f452a895fc54b40c94ef6 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22b5ad0dd53245948f996f0a02f34e50 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b31a93fa3244028a94ffe6c1f5eb09b : SUCCESS in 46m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/297c58ceb8194175bbc96ea031db4804 : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/78f9bbeb21e54084bddddc6ff5853fa9 : SUCCESS in 1h 08m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6987f5ab59b54d7c87ecdc9c6ca9b05d : SUCCESS in 1h 18m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7eacb2e802be4322b9320222e2b1386a : SUCCESS in 1h 15m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/265d2a32cc90453595f1c70b87af0b94 : FAILURE in 48m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c23ee73f471c48b7b0da7894b24d63eb : SUCCESS in 1h 04m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/91a0562759aa470ea04b2f7bead228e1 : SUCCESS in 1h 29m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/587914cba02745bbb2e7fcd43572a3a6 : SUCCESS in 1h 58m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04224b518585453ea630ad45de62a911 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":15},{"id":"1c2075665351098ec5b0154ecd0a6ab43a2dee9c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-03 00:40:06.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/60/786260/15/check/DellEMC_PowerMAX_FC/1addab0/DellEMC_PowerMAX_FC/303 : SUCCESS in 4h 15m 55s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/60/786260/15/check/DellEMC_PowerMAX_iSCSI/d721931/DellEMC_PowerMAX_iSCSI/296 : SUCCESS in 4h 39m 20s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":15},{"id":"ab37d2ee7db73846df39897cdc2bd39259f55416","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-03 03:47:31.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/15/check/dsvm-tempest-cisco-zonemanager-job/84e2b25/ : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":15},{"id":"2fdb5c72f0f4ee3d00303a42e50a12a7ceb2ea74","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-08-03 14:16:43.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"d95d8efc2f9d3c9fdf2c13c4c5a1172513a224d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 14:42:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d2493d5eb5145c38522c28ceae10f9b : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd595f37dfbe4e3ba905346e2b4cd9ed : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"738486e1f2cb0bd8d6e89c6530fae76a3d081210","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-03 15:11:13.000000000","message":"Patch Set 15:\n\nThe grenade failure is caused by https://bugs.launchpad.net/grenade/+bug/1938763\n\nThe fix is in the gate but has not merged yet: https://review.opendev.org/c/openstack/grenade/+/803317","accounts_in_message":[],"_revision_number":15},{"id":"dc6e25b0d83b22ec1ecd7a1af4257c13615cb013","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-03 15:13:38.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f740237e6da84ec49ef2fa019d21b08d : DISK_FULL in 45m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"f29a62d8978f0f4ca154b220f2e31dc93b22756f","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-08-03 15:22:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/0471cd6/index.html : SUCCESS in 1h 01m 33s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/89fb432/index.html : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":15},{"id":"f9cf18182a16e9390516b7b225984544d4fd66d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-03 15:59:47.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/786260/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d0a3c1f/ : SUCCESS in 1h 24m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/786260/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5da5900/ : SUCCESS in 1h 41m 33s","accounts_in_message":[],"_revision_number":15},{"id":"e872929c606067d79f619a98eefcf0f064ecb749","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-03 16:31:33.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/15/iscsi/ : FAILURE in 1h 32m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/15/smb/ : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":15},{"id":"2a68ccd0d84ca9319832efc5e2e747433f46aa35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 16:37:36.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a434e344f10491786cdc49fee009a1d : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcfe5eeefb0443ae98501f4e6e16be6a : SUCCESS in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c99dae981f1464f94b13e57ee5d970f : SUCCESS in 11m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def8ac670328434682264765d3c5742d : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3425b62ab16c40309123e220b23d8308 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/9036d1d543504a1f8cde266e6a065e78 : FAILURE in 48m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23d703380d064605b8c3c625097c6887 : SUCCESS in 1h 09m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9099680e23454d67be7645d24522b44f : SUCCESS in 7m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0263d22deb0425a91de5f960545d963 : SUCCESS in 12m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd293878fa85420aaa757e27465cc74c : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fda2351178c7492a9a4cd56dd9a8209b : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/909e60d44d56464da45aa16a05d1f9f2 : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1f5312dab494e52bc5dfd04ed0a67bc : SUCCESS in 44m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc285f0797684ffa8c88fb203e12c937 : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4dc4b7f6ac944428bfbeeb20b438361f : SUCCESS in 1h 19m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8ec44beb5ab54f7d8d4026c555418189 : SUCCESS in 1h 24m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f77929b65604474bb8379a12aab7cf9c : SUCCESS in 1h 17m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df97262d15fa4ea09ce1e833499448ff : FAILURE in 55m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/07a94392140c4fe889a9ffe87486093f : SUCCESS in 1h 26m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4f0b0c0e56854f9c873f9ea7f08b14f4 : SUCCESS in 1h 22m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f56e1f7c0f314dcd84a9792271353d2e : SUCCESS in 2h 14m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb1a8a6bc6124ad7b1757e4830289e04 : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":15},{"id":"e1f1d10c66fd694f515e50e5999ba158ac5087e8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-03 16:56:46.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/15/check/dsvm-tempest-cisco-zonemanager-job/baba928/ : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":15},{"id":"a54c6592043e0fdf6b67b2dd193ce5d034c01db3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-03 19:32:39.000000000","message":"Patch Set 15:\n\nrecheck grenade","accounts_in_message":[],"_revision_number":15},{"id":"abf9f530496235a614776e8f2df4cd8a71e448c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 19:56:05.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e5f551fbe0a478b8e9b6db2f2ef87fe : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67718ffb9ba64a59b68fcd7d8864d550 : SUCCESS in 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f09547ba0252a7d65600d5000180e9b46e8718b4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-03 21:23:51.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/786260/15/check/dsvm-tempest-cisco-zonemanager-job/5b74f70/ : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":15},{"id":"3989fedb159a358aa0071ead7c12ab3b412801ca","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-03 21:26:21.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786260/15/iscsi/ : SUCCESS in 1h 14m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/786260/15/smb/ : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":15},{"id":"39c0b71e1d40f54d422839767c090395a32d2456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 21:31:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08edc15ef094a578ba5513b69a64142 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/987390deb95f420d91385d50e2d3e7c8 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c4cdfc4fffd46eaae900de064a606cc : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1937a8a42eba49feaf217fc81f2997a9 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a458dc9d8f4774bda143de4950efa0 : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/8d35a3976e0b454dbfd52647ebb5f975 : SUCCESS in 56m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cdddc5952bfb49d7ba9d5acf9e5b7990 : SUCCESS in 1h 07m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73e6e02163594901beaf306eaed0443a : SUCCESS in 6m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/42c56bc5fb6e42288251f96949d884a7 : SUCCESS in 12m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8392c35bb6684bfcbeff048e96ad0d94 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cc105d4405145f2b8f49af1f6b28099 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c9ef5ff25d54613b074d0d1a096cc96 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7512ecf5f3e64d1d88b1069b82cb06d0 : SUCCESS in 36m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acbc6d4ba8714a6e96ecd02e19b734a7 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/152ae07a37fb466391531b256663bb7f : SUCCESS in 1h 08m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/529c09f3edbb4563a92f2f0186e7662e : SUCCESS in 1h 21m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d19e13a944c7411db3ce03d06eb6d8d6 : SUCCESS in 1h 25m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c89aa0afccf1444c9fb07eb7c02858b1 : SUCCESS in 1h 10m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b3b0c9554cd5401fb7e6870aa19792e5 : SUCCESS in 1h 11m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78e301a0e2d6472fbffe97ea638cb952 : SUCCESS in 1h 11m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9bb29084d69c4ede934220623f9a9a05 : SUCCESS in 1h 56m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2ae852b55fa4eee9545ea97c30f0cb3 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":15},{"id":"2e06d2cfc37674fa166dc6e39617589377ebddf8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-03 21:48:29.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/15/check/tempest-dsvm-full-focal-py3/f118773/ : FAILURE in 2h 03m 29s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/786260/15/check/devstack-plugin-ceph-dsvm-focal-py3/fbfd633/ : FAILURE in 52m 11s","accounts_in_message":[],"_revision_number":15},{"id":"faf731dc41b03be1df1f02054017d25ea3d43b65","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 22:09:36.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786260/786260/15/check-not-vote/ext-cinder-zuul/b6449c2 : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":15},{"id":"24f204f317fee7fbe10efa20f384fa6ef653600c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-03 22:31:33.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/786260/15/upstream-check/cinder-cDOT-iSCSI/01a9004/ : SUCCESS in 33m 18s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/786260/15/upstream-check/cinder-cDOT-NFS/1fb1306/ : FAILURE in 13m 04s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/786260/15/upstream-check/cinder-cDOT-FCP/4086a6c/ : SUCCESS in 59m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"e9cb46f665e3eb14c8e554d9c937cad984eddedf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-03 22:32:40.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/60/786260/15/check/DellEMC_PowerStore_iSCSI/eb4a978/DellEMC_PowerStore_iSCSI/280 : SUCCESS in 1h 00m 59s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/60/786260/15/check/DellEMC_PowerStore_FC/635dffb/DellEMC_PowerStore_FC/304 : SUCCESS in 47m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":15},{"id":"811cd2760207d79acf48af613d90ba476ea114c6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-03 22:49:28.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/60/786260/15/check/DellEMC_SC_FC/c41dd1d/DellEMC_SC_FC/221 : SUCCESS in 51m 41s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/60/786260/15/check/DellEMC_SC_iSCSI/b5dd76e/DellEMC_SC_iSCSI/217 : SUCCESS in 1h 17m 57s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"d2dfec9e8c0707c1e1bb17a919c7c7961785ee5c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-03 23:10:02.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/60/786260/15/check/DellEMC_PowerFlex/0243d5d/DellEMC_PowerFlex/338 : SUCCESS in 1h 38m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"73b09a5b67eb53a94059e98dfc18d42309c617d8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-03 23:14:14.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/786260/15/upstream-solidfire-check/cinder-solidfire-iSCSI/c982f82/ : FAILURE in 1h 41m 44s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"2f9301782079be95c650537945492c3d46f9405b","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-03 23:23:21.000000000","message":"Patch Set 15:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786260-15/Seagate-iSCSI : SUCCESS in 40:57.42\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786260-15/Seagate-FC : SUCCESS in 38:29.47","accounts_in_message":[],"_revision_number":15},{"id":"67dde470f63db2dca8b0f7ec8f99f56a9cc8b588","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-03 23:42:41.000000000","message":"Patch Set 15:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786260-15/HPMSA-iSCSI : FAILURE in 1:01:25\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786260-15/HPMSA-FC : SUCCESS in 34:04.77","accounts_in_message":[],"_revision_number":15},{"id":"6a2e76125a4cb63ed54da5985de57a835b12a13b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-04 00:27:35.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/786260/15/check/DellEMC_XtremIO_FC/1b2cce2/DellEMC_XtremIO_FC/586 : SUCCESS in 1h 19m 07s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/786260/15/check/DellEMC_XtremIO_iSCSI/f628500/DellEMC_XtremIO_iSCSI/572 : SUCCESS in 2h 02m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"30fba991b1935d7a4c5a19139633fb8750f22a00","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-04 00:28:25.000000000","message":"Patch Set 15:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786260-15/PVME-iSCSI : FAILURE in 1:35:08\n* PVME-FC http://os-logs.tristero.net/dell-ci/786260-15/PVME-FC : SUCCESS in 47:10.37","accounts_in_message":[],"_revision_number":15},{"id":"938c7821bf45f9876c1384cee15c16fae9f8818d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-04 02:12:07.000000000","message":"Patch Set 15:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786260-15/Lenovo-iSCSI : SUCCESS in 1:44:27\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786260-15/Lenovo-FC : SUCCESS in 2:25:22","accounts_in_message":[],"_revision_number":15},{"id":"605cd91e1f8ffe4bef68fa0a997dd798b9f48cf7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-04 12:51:07.000000000","message":"Patch Set 15: Code-Review+2 Review-Priority+1\n\nRenewing my +2 after the merge conflict resolution.\n\nRepeating from my earlier comment:\n\nThanks for the detailed commit message.  Code and tests LGTM.\n\nRaising the priority because this has come up on the ML and people are encountering this issue in real life:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-July/023784.html","accounts_in_message":[],"_revision_number":15},{"id":"e4429d3c404ef66be5e84ef558e3ca5412de916e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-05 05:17:13.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/60/786260/15/check/dsvm-tempest-TOYOU-ACS5000-driver/d3f1dd0 : FAILURE in 45m 57s","accounts_in_message":[],"_revision_number":15},{"id":"2fda5496790550d2655127e56865c404ec23ae94","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-09 15:01:28.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\nAlso renewing my vote here, thanks Jon for working on this. Looks good to me.","accounts_in_message":[],"_revision_number":15},{"id":"80235f53ddafc6522e34bd56813a7fd5642f0354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:02:04.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"864ec3172ef142773cb52e649d54f7197467ea6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:21:42.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e872681a603f43e59de679035e595af0 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faff4a1fe6ce4946a88599a19a924107 : SUCCESS in 10m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea23bd62496141a5b4bd23ea40df34e6 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6491d3852b554c539abf7cde482aae21 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/9fc6f33d66e74d86b037af36eb59f6f9 : SUCCESS in 1h 02m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbdf30c66cc447dba528db896747cd2d : SUCCESS in 59m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/682ebb6c79bd4e97a8f171bf1717bca1 : SUCCESS in 8m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9929ca03a7c43f0868b321969f1249e : SUCCESS in 1h 14m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c67287dfd7034a0d91951a2ef06cc8e4 : SUCCESS in 1h 13m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0895fcd17ce849bcb5ecd92c538c8525 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":15},{"id":"600683371e8b101b54cf06ed2e2bba32bc83ef9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:21:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"317ca0ddf13f4c4a1adcc303fff2cca8ccca206e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:23:21.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6331192b7be5459ebde429408a5227db : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c24efd4d1b3a4bedabfedfeda774d0a0 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"49a2c85eda9fd3cddc75fd904fe62c87a6b50735","revisions":{"3c300709aed69dcd0531bef9289ad76dd930e7e4":{"kind":"REWORK","_number":1,"created":"2021-04-14 15:15:55.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/1"}}},"commit":{"parents":[{"commit":"c12375fec6c005ba8d7e14d6f94612e7f5e66dd2","subject":"Merge \"Backup manager: Synchronously call remove_export_snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12375fec6c005ba8d7e14d6f94612e7f5e66dd2"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c300709aed69dcd0531bef9289ad76dd930e7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c300709aed69dcd0531bef9289ad76dd930e7e4"}]},"branch":"refs/heads/master"},"587e18ab97034ace16b52c7975e1459da0ca8b9c":{"kind":"REWORK","_number":2,"created":"2021-04-14 15:17:42.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/2"}}},"commit":{"parents":[{"commit":"c12375fec6c005ba8d7e14d6f94612e7f5e66dd2","subject":"Merge \"Backup manager: Synchronously call remove_export_snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c12375fec6c005ba8d7e14d6f94612e7f5e66dd2"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:17:23.000000000","tz":-240},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/587e18ab97034ace16b52c7975e1459da0ca8b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/587e18ab97034ace16b52c7975e1459da0ca8b9c"}]},"branch":"refs/heads/master"},"a15e4d67b3f27dd2c5a8cffe37cab8e118957837":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2021-06-06 11:11:57.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/3"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-06-06 11:11:39.000000000","tz":-240},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a15e4d67b3f27dd2c5a8cffe37cab8e118957837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a15e4d67b3f27dd2c5a8cffe37cab8e118957837"}]},"branch":"refs/heads/master"},"9ccfb22dbce6369810fd360b461dcbe1a0c515aa":{"kind":"REWORK","_number":4,"created":"2021-06-10 18:07:59.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/4"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-06-10 18:07:35.000000000","tz":-240},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ccfb22dbce6369810fd360b461dcbe1a0c515aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ccfb22dbce6369810fd360b461dcbe1a0c515aa"}]},"branch":"refs/heads/master"},"da19a11478e17605d2a12cdfe736b6da0fd10805":{"kind":"REWORK","_number":5,"created":"2021-06-10 23:32:39.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/5"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-06-10 23:32:28.000000000","tz":-240},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da19a11478e17605d2a12cdfe736b6da0fd10805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da19a11478e17605d2a12cdfe736b6da0fd10805"}]},"branch":"refs/heads/master"},"e4eecf6e7b6d01a48d1525e01f403d3e7558fff9":{"kind":"REWORK","_number":6,"created":"2021-06-11 18:19:19.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/6"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-06-11 18:19:08.000000000","tz":-240},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of the \u0027order\u0027 parameter would cause an error to be\nreturned where previous versions would correct the value at runtime.\nThis becomes a problem when creating a volume from an image, where the\nsource RBD image may have a larger stripe unit than cinder\u0027s RBD driver\nis configured for.  When this happens, clone() is called with an order\nthat is too small given that of the source image and the clone fails.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values (used to compute order) so that the\ndesired stripe unit is respected.\n\nAdditionally, we are also seeing this bug because Glance\u0027s default\nstripe width is different (larger) from Cinder\u0027s.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4eecf6e7b6d01a48d1525e01f403d3e7558fff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4eecf6e7b6d01a48d1525e01f403d3e7558fff9"}]},"branch":"refs/heads/master"},"b887cce3b028c57536d3c14066335029f4a4f9ca":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-06-21 03:45:11.000000000","uploader":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"ref":"refs/changes/60/786260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/7"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"jiang guodao","email":"jguodao@vmware.com","date":"2021-06-21 03:45:11.000000000","tz":0},"subject":"RBD: use correct stripe size in clone operation","message":"RBD: use correct stripe size in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of the \u0027order\u0027 parameter would cause an error to be\nreturned where previous versions would correct the value at runtime.\nThis becomes a problem when creating a volume from an image, where the\nsource RBD image may have a larger stripe unit than cinder\u0027s RBD driver\nis configured for.  When this happens, clone() is called with an order\nthat is too small given that of the source image and the clone fails.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values (used to compute order) so that the\ndesired stripe unit is respected.\n\nAdditionally, we are also seeing this bug because Glance\u0027s default\nstripe width is different (larger) from Cinder\u0027s.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b887cce3b028c57536d3c14066335029f4a4f9ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b887cce3b028c57536d3c14066335029f4a4f9ca"}]},"branch":"refs/heads/master","description":"Rebase"},"07f6172697c8e92d29f2bab53a2e83661d316232":{"kind":"REWORK","_number":8,"created":"2021-07-25 21:28:50.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/8"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-25 21:17:57.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default strip unit of 8MB (distinctly larger than\nthat of Cinder).  If we do not consider the incoming strip unit and\nselect the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values so that the appropriate stripe unit\nis chosen.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07f6172697c8e92d29f2bab53a2e83661d316232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07f6172697c8e92d29f2bab53a2e83661d316232"}]},"branch":"refs/heads/master"},"99043b883b601e6922850660c9afccb55cbe336b":{"kind":"REWORK","_number":9,"created":"2021-07-25 21:36:44.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/9"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-25 21:36:28.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default strip unit of 8MB (distinctly larger than\nthat of Cinder).  If we do not consider the incoming strip unit and\nselect the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values so that the appropriate stripe unit\nis chosen.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99043b883b601e6922850660c9afccb55cbe336b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99043b883b601e6922850660c9afccb55cbe336b"}]},"branch":"refs/heads/master"},"8ff113d79b749c7256abf6493e510815c6a6f193":{"kind":"REWORK","_number":10,"created":"2021-07-25 22:05:24.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/10"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-25 22:05:13.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default strip unit of 8MB (distinctly larger than\nthat of Cinder).  If we do not consider the incoming strip unit and\nselect the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values so that the appropriate stripe unit\nis chosen.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ff113d79b749c7256abf6493e510815c6a6f193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ff113d79b749c7256abf6493e510815c6a6f193"}]},"branch":"refs/heads/master"},"dcd67ce7a658bde78dadfeb792df3dc169b52a9d":{"kind":"REWORK","_number":11,"created":"2021-07-26 15:23:17.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/11"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-26 15:23:05.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default strip unit of 8MB (distinctly larger than\nthat of Cinder).  If we do not consider the incoming strip unit and\nselect the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values so that the appropriate stripe unit\nis chosen.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcd67ce7a658bde78dadfeb792df3dc169b52a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcd67ce7a658bde78dadfeb792df3dc169b52a9d"}]},"branch":"refs/heads/master"},"9657f31326133a93278ba74a05b5113e3e8145b6":{"kind":"REWORK","_number":12,"created":"2021-07-26 18:58:42.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/12"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-26 18:57:15.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default strip unit of 8MB (distinctly larger than\nthat of Cinder).  If we do not consider the incoming strip unit and\nselect the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values so that the appropriate stripe unit\nis chosen.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9657f31326133a93278ba74a05b5113e3e8145b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9657f31326133a93278ba74a05b5113e3e8145b6"}]},"branch":"refs/heads/master"},"f58262290538e99743bb5a98e77e5e588270108d":{"kind":"REWORK","_number":13,"created":"2021-07-27 17:17:48.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/13"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-27 17:17:32.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default strip unit of 8MB (distinctly larger than\nthat of Cinder).  If we do not consider the incoming strip unit and\nselect the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two strip unit values so that the appropriate stripe unit\nis chosen.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f58262290538e99743bb5a98e77e5e588270108d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f58262290538e99743bb5a98e77e5e588270108d"}]},"branch":"refs/heads/master"},"a022348fd519987337df16767eace970a794e5ea":{"kind":"REWORK","_number":14,"created":"2021-07-28 02:43:28.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/14"}}},"commit":{"parents":[{"commit":"c2f9c84aa6960651790bd0f5d68185bbd58708d4","subject":"Merge \"NFS: Update connection info on online snap create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2f9c84aa6960651790bd0f5d68185bbd58708d4"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-07-28 02:38:41.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default stripe unit of 8MB (distinctly larger\nthan that of Cinder).  If we do not consider the incoming stripe unit\nand select the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two stripe unit values so that the appropriate stripe unit\nis chosen.\n\nIt should also be noted that we\u0027re determining the correct stripe unit,\nbut using the \u0027order\u0027 argument to clone().  Ceph will set the stripe\nunit equal to the object size (order) by default and we rely on this\nbehaviour for the following reason: passing stripe-unit alone or with\nan order argument causes an invalid argument exception to be raised in\npre-pacific releases of Ceph, as it\u0027s argument parsing appears to have\nlimitations.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a022348fd519987337df16767eace970a794e5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a022348fd519987337df16767eace970a794e5ea"}]},"branch":"refs/heads/master"},"49a2c85eda9fd3cddc75fd904fe62c87a6b50735":{"kind":"REWORK","_number":15,"created":"2021-08-02 15:13:45.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/60/786260/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/786260/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/786260/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/786260/15"}}},"commit":{"parents":[{"commit":"b06a300eac9726a1589fd670ce0617d2ff324870","subject":"Merge \"[SVF] Fixed add volumes to clone group issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b06a300eac9726a1589fd670ce0617d2ff324870"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-04-14 15:14:13.000000000","tz":-240},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2021-08-02 15:13:13.000000000","tz":-240},"subject":"RBD: use correct stripe unit in clone operation","message":"RBD: use correct stripe unit in clone operation\n\nThe recent release of Ceph Pacific saw a change to the clone() logic\nwhere invalid values of stripe unit would cause an error to be returned\nwhere previous versions would correct the value at runtime.  This\nbecomes a problem when creating a volume from an image, where the source\nRBD image may have a larger stripe unit than cinder\u0027s RBD driver is\nconfigured for.  When this happens, clone() is called with a stripe unit\nthat is too small given that of the source image and the clone fails.\n\nThe RBD driver in Cinder has a configuration parameter\n\u0027rbd_store_chunk_size\u0027 that stores the preferred object size for cloned\nimages.  If clone() is called without a stripe_unit passed in, the\nstripe unit defaults to the object size, which is 4MB by default.  The\nissue arises when creating a volume from a Glance image, where Glance is\ncreating images with a default stripe unit of 8MB (distinctly larger\nthan that of Cinder).  If we do not consider the incoming stripe unit\nand select the larger of the two, Ceph cannot clone an RBD image with a\nsmaller stripe unit and raises an error.\n\nThis patch adds a function in our driver\u0027s clone logic to select the\nlarger of the two stripe unit values so that the appropriate stripe unit\nis chosen.\n\nIt should also be noted that we\u0027re determining the correct stripe unit,\nbut using the \u0027order\u0027 argument to clone().  Ceph will set the stripe\nunit equal to the object size (order) by default and we rely on this\nbehaviour for the following reason: passing stripe-unit alone or with\nan order argument causes an invalid argument exception to be raised in\npre-pacific releases of Ceph, as it\u0027s argument parsing appears to have\nlimitations.\n\nCloses-Bug: #1931004\nChange-Id: Iec111ab83e9ed8182c9679c911e3d90927d5a7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49a2c85eda9fd3cddc75fd904fe62c87a6b50735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49a2c85eda9fd3cddc75fd904fe62c87a6b50735"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
