)]}'
{"id":"openstack%2Fcinder~786926","triplet_id":"openstack%2Fcinder~master~Ia2b27874a2c35e7837b6d61dc4cb81d5ada5d5ae","project":"openstack/cinder","branch":"master","topic":"bp/alembic-migration","hashtags":[],"change_id":"Ia2b27874a2c35e7837b6d61dc4cb81d5ada5d5ae","subject":"db: Compact migrations to Train","status":"MERGED","created":"2021-04-19 16:57:34.000000000","updated":"2021-06-25 22:13:33.000000000","submitted":"2021-06-23 23:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786926-bp/alembic-migration","meta_rev_id":"5ebffdca012731e510a62f927a0c2a2547c667ab","_number":786926,"virtual_id_number":786926,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-23 23:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-05-04 20:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-23 13:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-23 14:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-19 17:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-19 18:07:11.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-19 19:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 19:42:09.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-19 20:47:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-19 23:54:10.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 03:18:46.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 04:42:58.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 06:16:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 18:54:26.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-20 21:27:33.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 22:01:59.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-20 22:01:59.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-20 22:56:30.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-20 23:05:58.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-20 23:28:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-21 01:17:06.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-21 04:54:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 10:36:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 14:50:29.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-22 05:32:48.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 20:58:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-23 13:38:39.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2021-06-24 14:22:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"75136af0d9cb259bacdbe063ece0455b1c2bbd25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-19 16:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a28a73c54692a1f51a341716a5f06f1b9271cc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 17:42:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c351dcc2fe04649b3692afb79bc7bb6 : SUCCESS in 30m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3eebd9b7c0224b2ebe49c3ad4dd7e1da : SUCCESS in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2b45d4aafffea30fc0e939582799886e08287c5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-19 18:07:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786926/1/iscsi/ : FAILURE in 53m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/786926/1/smb/ : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"40866bc724c4efe0c79e559e8372f71f533aaac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 19:36:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/547dfa001f1046efa5370d5da38de684 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f183410121b849b789e195415ee756ff : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf6eaaf59bed49d98fc8309534eaaab8 : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bda014c7adc1452dbe7dd25b3a895d15 : SUCCESS in 13m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/420256ecd133486f8512bd88e7547bd1 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d5bdd95f3a47b0b2991a9edd1bcc55 : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/d500ec87ed0642b7883561c9317dfd87 : SUCCESS in 1h 05m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1dab2b6e5a874022bc4e7b301bef524d : SUCCESS in 1h 15m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cd9480bbaa454fdfb213fafa7702eed2 : SUCCESS in 13m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b672beef2234496fae344aec12067b3c : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8e7fb9b4f3143768dc5f62d2f4482e8 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4eec58af0c584834b459290bcc687fa8 : SUCCESS in 8m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4355485157664bd781799dce82cc38de : SUCCESS in 36m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9081ed3a8ef046029e2069463c5ddf33 : FAILURE in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6444de7c5eb43608eb9c915fc4d08d1 : SUCCESS in 1h 11m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b23bc983d216425ca1efb5885a8942f0 : FAILURE in 1h 13m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02577f98338f499b801f828ff8335298 : FAILURE in 1h 52m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71428739f7a04d5bb99bd8ce662a7d32 : SUCCESS in 1h 30m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0623e78abf78433c8540c9df580b6bad : SUCCESS in 1h 23m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7b5d7327d4684e119567161fb81d974b : SUCCESS in 1h 18m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72dee4fd1a5341f8b7393b5e960a4faf : SUCCESS in 2h 02m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c4fc5aa93dd4c1baa9f430d8b978116 : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a3fd442dab8b0190dea43326bab8c53c2809dadd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-19 19:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/26/786926/1/check/dsvm-tempest-cisco-zonemanager-job/e7e253a/ : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fe0efc0b68ab9a879637ed87023bd90f24b61b0b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-19 20:47:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/26/786926/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a1c5b26/ : SUCCESS in 2h 20m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/26/786926/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ffdd6c7/ : SUCCESS in 2h 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"82b52a0eee5dee04340615fce8a8573b964cb3f6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-19 23:54:10.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-26-786926-1/ns5_nfs : SUCCESS in 5h 1m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-26-786926-1/ns5_iscsi : SUCCESS in 4h 53m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"37dc5c37bb603bdc3c267f54a92be3e0e741ff37","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 03:18:46.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/26/786926/1/check/DellEMC_XtremIO_FC/9634c32/DellEMC_XtremIO_FC/9 : SUCCESS in 2h 05m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/26/786926/1/check/DellEMC_XtremIO_iSCSI/32859c9/DellEMC_XtremIO_iSCSI/9 : SUCCESS in 1h 54m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"2cba60d92e5429a5febd1ded3f70d2b042499f73","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 04:42:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/26/786926/1/check/iscsi-driver-nimble/1b5ac68/1b5ac68.html : FAILURE in 1h 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0f39fea6aec56ffab49c669b6ef157b9a66c86d5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 06:16:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/786926/1/20210420103600/NEC-ISCSI : SUCCESS in 2h 11m 46s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/786926/1/20210420103600/NEC-FC : SUCCESS in 4h 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6196237deac593fe52e724caf21bc1e2a4284862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-20 16:35:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"81070c9be3cad79eee5f3669b1e1683a195fa56d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:21:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7110ad00bd7a4bcdb2c99362bdd3dc2c : SUCCESS in 29m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c395746ff2b14b0fb2030c66f536f630 : FAILURE in 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9386c5a3be4b55eabbae6d0f3f17364e53f95832","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-20 17:36:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786926/2/iscsi/ : SUCCESS in 58m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/786926/2/smb/ : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6e6f700b11f67fb0e1625e448648445b93a02e2e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 18:02:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/26/786926/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/50e1004/50e1004.html : FAILURE in 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"dc6b938a8fbe6a5d52222358a1b6757e425ebd58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 18:42:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b56784992df6404589190e1e9e97d53f : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f23b56b61a914a748a73efeb962336a9 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7520902b241a4ac1844b3ec1a2b6cda8 : SUCCESS in 13m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4bfd3b79a7b4b1e8f7491485772bc57 : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bcbe61b6af94b32bf3d16b872f5893e : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d925cc1a11246199c70de38e4c32fc0 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/00a6bfa97d2545a7b3476772421b737b : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23ec4b36f417484da071a49e1fce044a : SUCCESS in 1h 00m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21fa462f7ba541fab289bc10830eed00 : SUCCESS in 15m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3b671b12bea34bf7b792b0fc4f8bf1e1 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3d734832b054ec59ff56bb12ca0e294 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e766a583d48142938f54c0a17f44039d : SUCCESS in 8m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/790aa26f239e4e098e3af5b94b3450a9 : SUCCESS in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d1120bd90f84834ab5a4ef2e0bc648a : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e70284a2f66f4d74adf488d7ab949c5a : SUCCESS in 1h 12m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0594db157e94470599fade53cc7a97de : FAILURE in 1h 31m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d4295219cb348b7a96a54434858f99f : SUCCESS in 1h 34m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6324d4c009bf438aa48d96141afb1e9e : SUCCESS in 1h 19m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0acbdd9e195c49a7ab6bf6a0874563f4 : SUCCESS in 1h 17m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5f928935c3124da6a59eaa2ae87e6010 : SUCCESS in 1h 36m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d9b956174274bba9b5c7fc61a6a9d28 : SUCCESS in 1h 58m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0566fdd2a884f319af4323fb1476bf4 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b217b198b2b4afcee133a2f62431fb5bef17c81f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-20 18:54:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/26/786926/2/upstream-solidfire-check/cinder-solidfire-iSCSI/c9a365c/ : FAILURE in 10m 08s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"97c567107bcfb3efe6da4a502994bfae03797112","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-20 19:14:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/26/786926/2/check/dsvm-tempest-cisco-zonemanager-job/f41c03f/ : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0da922fed40213fbe2bce45045220f0f3925e233","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-20 19:26:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/26/786926/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9420918/ : SUCCESS in 2h 26m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/26/786926/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/82183d5/ : SUCCESS in 2h 08m 00s","accounts_in_message":[],"_revision_number":2},{"id":"606addc54bbd69a93b1acd9067d615222fd79234","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-20 21:27:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/26/786926/2/upstream-check/cinder-cDOT-iSCSI/f93aea3/ : SUCCESS in 35m 05s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/26/786926/2/upstream-check/cinder-cDOT-NFS/d484419/ : SUCCESS in 2h 27m 55s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/26/786926/2/upstream-check/cinder-cDOT-FCP/387ad26/ : FAILURE in 57m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5fb8cde6138f902f5b21ad3e396a51ae5e79c085","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-20 22:01:59.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786926-2/HPMSA-iSCSI : FAILURE in 59:22.61\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786926-2/HPMSA-FC : SUCCESS in 33:28.59","accounts_in_message":[],"_revision_number":2},{"id":"edb759fe698833b6e65f8fe5f65b44adb7d9f8b0","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-20 22:01:59.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786926-2/Lenovo-iSCSI : SUCCESS in 48:34.06\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786926-2/Lenovo-FC : SUCCESS in 45:22.75","accounts_in_message":[],"_revision_number":2},{"id":"16995f2d9a1a73d40794beadad4408e86e01a0f5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 22:20:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/26/786926/2/check/iscsi-driver-primera/ee7b4f7/ee7b4f7.html : FAILURE in 1h 05m 41s","accounts_in_message":[],"_revision_number":2},{"id":"789147519de3aca64d1c6db3649752dd0252d17c","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-20 22:56:30.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786926-2/PVME-iSCSI : SUCCESS in 1:00:43\n* PVME-FC http://os-logs.tristero.net/dell-ci/786926-2/PVME-FC : SUCCESS in 52:53.30","accounts_in_message":[],"_revision_number":2},{"id":"a237f0c9765ed5784624b581de04ddb6a5427adc","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-20 23:05:58.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786926-2/Seagate-iSCSI : SUCCESS in 46:32.22\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786926-2/Seagate-FC : SUCCESS in 39:31.33","accounts_in_message":[],"_revision_number":2},{"id":"d351942784dbd9f5c489f8d29e1902617d9aa36d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-20 23:28:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_Unity_FC/10e53fb/DellEMC_Unity_FC/13 : SUCCESS in 1h 29m 49s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_Unity_iSCSI/74a409b/DellEMC_Unity_iSCSI/13 : SUCCESS in 1h 33m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"57390765d907dabab68f5da7f0cc1dec9cca028c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-20 23:30:19.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-26-786926-2/ns5_nfs : SUCCESS in 4h 55m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-26-786926-2/ns5_iscsi : SUCCESS in 4h 57m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"39d7817adc61b0303387ad48d2ff77a347fbb06d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-21 01:17:06.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_SC_FC/f0a2ef8/DellEMC_SC_FC/13 : SUCCESS in 1h 55m 34s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_SC_iSCSI/275c1b5/DellEMC_SC_iSCSI/12 : SUCCESS in 2h 02m 33s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"411d70b0cf663bdbf0ae32a9d2694e5f579effe0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-21 02:47:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/26/786926/2/check/iscsi-driver-nimble/fc8b739/fc8b739.html : SUCCESS in 1h 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4b361d9349edec502b1239886db1ab11bbf8840d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 04:54:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/26/786926/2/check/DellEMC_PowerFlex/3860bbf/DellEMC_PowerFlex/12 : SUCCESS in 2h 33m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"e3ee667c1209ab1d68686101a7b284469f9fc362","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 10:36:03.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/786926/2/check/tempest-dsvm-full-focal-py3/05781f2/ : FAILURE in 51m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/945fd557d6a142299dc4cd5d3a561fc1 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a43672338133598a5f8130d136a8280f4fe5c6e3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 14:50:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_VNX_iSCSI/d7e9740/DellEMC_VNX_iSCSI/12 : FAILURE in 2h 56m 24s\n- DellEMC_VNX_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_VNX_FC/7cb04a2/DellEMC_VNX_FC/14 : SUCCESS in 1h 50m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"c0a2cb7388767ddda4393c88be00b268310cd141","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 15:59:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_XtremIO_FC/d7823d2/DellEMC_XtremIO_FC/27 : FAILURE in 2h 06m 33s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_XtremIO_iSCSI/c119910/DellEMC_XtremIO_iSCSI/27 : SUCCESS in 1h 52m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3b677625595f1d4ac273012ff0ceebde19039e8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 16:12:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/786926/2/20210421222100/NEC-FC : SUCCESS in 2h 43m 04s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/786926/2/20210421222100/NEC-ISCSI : SUCCESS in 2h 51m 35s","accounts_in_message":[],"_revision_number":2},{"id":"968bd55b8323a8fbe952e123d567c1c6d3906540","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-22 05:32:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_PowerMAX_FC/a099cf5/DellEMC_PowerMAX_FC/9 : SUCCESS in 4h 31m 52s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_PowerMAX_iSCSI/ccb3553/DellEMC_PowerMAX_iSCSI/9 : SUCCESS in 5h 18m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f9f770dd3487efbbd9d882468e4ca63b00e1ee39","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 20:58:16.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"b411186eae5e77f87ad8de844eeadd0b4cef343e","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-06-23 13:38:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f0f2f39bdd26790d8e1934f745a8d3655b3ddc84","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-06-23 14:07:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fee046c3334e480fe920c92a05a9a92b8587dd32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 14:08:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d28e81e76cf6703a6eec0320a9602b49c4adac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:02:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/866b4c72691945ef8ce89226e1b0a207 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf4a45cb26148eda5c742f732e67dd9 : SUCCESS in 10m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/309f143e80564e7795e83b87bd1f127b : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/974bf9ea1c7246be98367e27526872dc : SUCCESS in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e094caec04d4c238049ec1bb9970528 : SUCCESS in 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/b891986c3d1b460abf0a09be5c60893a : FAILURE in 1h 12m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc3e5e8a18f949248dfb717f8b1c2245 : SUCCESS in 1h 09m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/61b02c91e89f47f7a56df931bf49fdc3 : SUCCESS in 1h 20m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e7f1751dfc3143f092b8e1d29e09eb82 : SUCCESS in 1h 39m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3263af44533c484ba11efb5b4cbb25c2 : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e3a8e5d7dea8dbe44fa5e897ffd3c6b9fa92c93c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-23 19:07:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7c0ba6024927103b1667f1d5c86c71fb19b4c510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:26:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08f857ba5b3f492b8560dbb294dadf86 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49417a13725a4345b5959f6120a67538 : SUCCESS in 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fe6baf4295c33c3ef6e8ad994971c3ff15de53d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-23 20:04:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786926/2/iscsi/ : SUCCESS in 56m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/786926/2/smb/ : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":2},{"id":"858f2c7b829dc540b4c0e4ad430b6c76114379e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:16:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8da3a33a6f4340f9b6bbe5fe17498c4b : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a341a8043434d5992d23e7c66e0ab26 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1afa7c42196f4a8bae6555bd8a3cb439 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b18bb4f9c694701bec5a3cb60bb0598 : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d251a879560e46db98de974c8bd26d92 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854824c15be948e9ad4bd1f5658ef8de : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/bd33ccdf522a426087505d05de0433e3 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e3c317642d347f8bb2d4f522c8a475c : SUCCESS in 1h 30m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a3321b242b444cceb8c24c226db72bbd : SUCCESS in 12m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80235286204f450f9615d10c7da88320 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d45b6c97c8340819811e60885ea0eb4 : SUCCESS in 8m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8bed2589594f46868d5af5947dfff975 : SUCCESS in 8m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b7081dd82592449e9835455b427ab018 : FAILURE in 30m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d62dd91961564676b0a91d13bff8e0ee : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94713b3b50504230a1b2665577a92a8b : SUCCESS in 1h 04m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a8e7c3a79044018ba7e59a71d11642b : SUCCESS in 1h 29m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66de7b13fbca41c5b8b524258c2a7a30 : SUCCESS in 1h 30m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/642c00109ef14158a6168d111f161714 : SUCCESS in 1h 16m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/36f4cfa8a751427ba1114ccfa39e72dc : SUCCESS in 1h 18m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5e282988bdd04345a28f0df0d140c7fa : SUCCESS in 1h 24m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88da020daa924409aa20d85da76bc78c : SUCCESS in 2h 07m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b92dce88ca7c4f10836566757d3dee02 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":2},{"id":"26b352feb0df2fe62da901064da3903f01a1246f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:16:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60e29a14d3af7e81c5fd3419a5f0a45382823e61","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-23 22:02:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/26/786926/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a9ca7a9/ : SUCCESS in 1h 39m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/26/786926/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a7d9dd8/ : SUCCESS in 1h 41m 00s","accounts_in_message":[],"_revision_number":2},{"id":"bd6bf4c106f3e6bf79ded5dde87750bc326f03ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:05:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ee4d3b561f04093a9bff74206600812 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d0260d79014a3e822cefc7ef82cd57 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aa6feb063b54b8596a8666b35dd9609 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da1cb0a716e342f5a670ade86c4c3f67 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0b41a715be407cb6ddf74571805164 : SUCCESS in 13m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8c0d0e97bb3e46ce9faf7ed13f182582 : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aba2127eeeac4aa39ee2747947a8cada : SUCCESS in 1h 01m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6db999d10a54db68aa743e8d650e21e : SUCCESS in 1h 13m 27s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1ebf405926c43ce90cd2771466dc36a : SUCCESS in 1h 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a78980390fdd480cac13efbb15c6ce6d : SUCCESS in 46m 58s","accounts_in_message":[],"_revision_number":2},{"id":"844fcd14f159f4c513be682e1aa1d67b394b285e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:05:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"40432a1298421bb048e8ab8d058a9d32f9620c95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:06:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70164cbc1c0e425b8ac11ae4efe57012 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"95c30713119cb219a3ca5e0bb5ed8ede3b864398","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-23 23:14:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/26/786926/2/check/dsvm-tempest-cisco-zonemanager-job/a4e75db/ : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2d2f97d17e28fd6159a66186d532b4210f958dbe","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-23 23:20:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/26/786926/2/upstream-check/cinder-cDOT-iSCSI/fef38b9/ : SUCCESS in 35m 00s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/26/786926/2/upstream-check/cinder-cDOT-NFS/85e5113/ : SUCCESS in 2h 03m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/26/786926/2/upstream-check/cinder-cDOT-FCP/470ac02/ : SUCCESS in 57m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f549259109ab6e5780256412c796f2bd12b0ba3a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-23 23:59:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/26/786926/2/upstream-solidfire-check/cinder-solidfire-iSCSI/8fe7c6e/ : SUCCESS in 2h 41m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ebd276af94dbfd4a1b0d19084d9b93d69ad0a77","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-24 00:40:42.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/786926/2/check/tempest-dsvm-full-focal-py3/d9ccc77/ : SUCCESS in 1h 45m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/786926/2/check/devstack-plugin-ceph-dsvm-focal-py3/0d11bac/ : FAILURE in 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f29301c36b12495870a1a11f17b98f5af8458452","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-06-24 11:23:07.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786926-2/Seagate-iSCSI : FAILURE in 40:54.37\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786926-2/Seagate-FC : FAILURE in 35:59.24","accounts_in_message":[],"_revision_number":2},{"id":"4a4f45050d78a412166237d9fa73564c2d444db6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 12:49:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/26/786926/2/check/DellEMC_PowerFlex/5f43e2b/DellEMC_PowerFlex/219 : SUCCESS in 1h 38m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"898f381d38a03eb32d577149c32d54da6efa647a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-24 14:22:47.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786926/786926/2/check-not-vote/ext-cinder-zuul/49ef649 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4b76622e0c36bba0f0edcace4d47b5e6f3829aac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-24 18:16:22.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_Unity_FC/7819831/DellEMC_Unity_FC/348 : SUCCESS in 1h 21m 34s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_Unity_iSCSI/db09fa8/DellEMC_Unity_iSCSI/328 : SUCCESS in 2h 14m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"31ef22f1697eaf3e710da5e72b3a4f6b451c3440","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-24 21:21:09.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786926-2/HPMSA-iSCSI : FAILURE in 55:36.68\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786926-2/HPMSA-FC : FAILURE in 30:43.37","accounts_in_message":[],"_revision_number":2},{"id":"52da77af29892bc812bfb848b4e4c156604a449e","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-24 21:35:33.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786926-2/Lenovo-iSCSI : FAILURE in 44:58.99\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786926-2/Lenovo-FC : FAILURE in 42:32.29","accounts_in_message":[],"_revision_number":2},{"id":"f373598a33d1cf6964e2081d0d907b06b987eab9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 22:28:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/26/786926/2/check/DellEMC_XtremIO_FC/0f84ae8/DellEMC_XtremIO_FC/397 : FAILURE in 1h 18m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/26/786926/2/check/DellEMC_XtremIO_iSCSI/c90f31c/DellEMC_XtremIO_iSCSI/392 : SUCCESS in 2h 22m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ebffdca012731e510a62f927a0c2a2547c667ab","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-25 22:13:33.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786926-2/PVME-iSCSI : FAILURE in 51:56.10\n* PVME-FC http://os-logs.tristero.net/dell-ci/786926-2/PVME-FC : FAILURE in 51:58.55","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b29bb44e98e00c68382b3f381a0a6e0938fe6dda","revisions":{"8e6b86b7c902a27e855708ffe2ffc4cc638dcef9":{"kind":"REWORK","_number":1,"created":"2021-04-19 16:57:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/786926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/786926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/786926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/786926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/786926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/786926/1"}}},"commit":{"parents":[{"commit":"f4ad89c1e9a784f8c8379c3a5992396bac517752","subject":"db: Remove \u0027cinder.database.migration_backend\u0027 entrypoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4ad89c1e9a784f8c8379c3a5992396bac517752"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 10:36:34.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 14:21:07.000000000","tz":60},"subject":"db: Compact migrations to Train","message":"db: Compact migrations to Train\n\nCompare all database migrations up to Train into the initial schema to\nspeed up migrations and make switching to alembic easier. Changes\ninclude:\n\n127_change_project_resource_attribute_for_quota_usages\n  Change type of \u0027quota_usages.resource\u0027 to CHAR(300)\n\n128_add_project_id_and_accepted_to_transfer\n  Add \u0027source_project_id\u0027, \u0027destination_project_id\u0027, and \u0027accepted\u0027\n  columns to \u0027transfers\u0027 table\n\n132_create_default_volume_type\n  Add default entry to \u0027volume_types\u0027 entry\n\nChange-Id: Ia2b27874a2c35e7837b6d61dc4cb81d5ada5d5ae\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e6b86b7c902a27e855708ffe2ffc4cc638dcef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e6b86b7c902a27e855708ffe2ffc4cc638dcef9"}]},"branch":"refs/heads/master"},"b29bb44e98e00c68382b3f381a0a6e0938fe6dda":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-20 16:35:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/786926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/786926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/786926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/786926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/786926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/786926/2"}}},"commit":{"parents":[{"commit":"59c5f1e7adad55002e1c80e6d9d3052dd8431d23","subject":"db: Remove \u0027cinder.database.migration_backend\u0027 entrypoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59c5f1e7adad55002e1c80e6d9d3052dd8431d23"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 10:36:34.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-20 16:13:22.000000000","tz":60},"subject":"db: Compact migrations to Train","message":"db: Compact migrations to Train\n\nCompare all database migrations up to Train into the initial schema to\nspeed up migrations and make switching to alembic easier. Changes\ninclude:\n\n127_change_project_resource_attribute_for_quota_usages\n  Change type of \u0027quota_usages.resource\u0027 to CHAR(300)\n\n128_add_project_id_and_accepted_to_transfer\n  Add \u0027source_project_id\u0027, \u0027destination_project_id\u0027, and \u0027accepted\u0027\n  columns to \u0027transfers\u0027 table\n\n132_create_default_volume_type\n  Add default entry to \u0027volume_types\u0027 entry\n\nChange-Id: Ia2b27874a2c35e7837b6d61dc4cb81d5ada5d5ae\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b29bb44e98e00c68382b3f381a0a6e0938fe6dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b29bb44e98e00c68382b3f381a0a6e0938fe6dda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
