)]}'
{"id":"openstack%2Fcinder~786927","triplet_id":"openstack%2Fcinder~master~Icb1cf73722e0e9bf1364d64522e38b94aa716bea","project":"openstack/cinder","branch":"master","topic":"bp/remove-sqlalchemy-migrate","hashtags":[],"change_id":"Icb1cf73722e0e9bf1364d64522e38b94aa716bea","subject":"db: Use \u0027import sqlalchemy as sa\u0027 pattern","status":"MERGED","created":"2021-04-19 16:57:34.000000000","updated":"2021-08-25 11:06:20.000000000","submitted":"2021-08-25 11:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786927-bp/remove-sqlalchemy-migrate","meta_rev_id":"d1c7475c244bc2228f741c6dd13e65981cecb512","_number":786927,"virtual_id_number":786927,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 11:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-25 11:06:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-07-16 07:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-08-04 15:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-25 09:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-25 09:40:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26159,"name":"xiexianbin","email":"xiexianbin@yovole.com","username":"xianbin"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-19 17:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-19 18:29:48.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-19 19:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 20:50:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-19 21:29:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-20 00:58:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 05:23:43.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 06:04:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 07:32:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 08:44:11.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 21:04:30.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-21 00:16:37.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-21 00:20:00.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-21 00:50:01.000000000","updated_by":{"_account_id":26159,"name":"xiexianbin","email":"xiexianbin@yovole.com","username":"xianbin"},"reviewer":{"_account_id":26159,"name":"xiexianbin","email":"xiexianbin@yovole.com","username":"xianbin"},"state":"CC"},{"updated":"2021-04-21 01:03:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-21 01:12:28.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-21 01:35:25.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-21 03:30:52.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-21 06:46:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 07:37:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 08:12:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-21 17:46:28.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-22 11:52:10.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 20:58:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-05-27 14:26:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-07-13 22:11:26.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 03:51:10.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2021-07-14 12:03:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-07-14 14:29:08.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 07:03:00.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-08-25 09:40:40.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"cfa0a9323cf87233f70268d7c4ac7bcd30721755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-19 16:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"becca84d3ccd0ee57d2050a971bf4715911d8eb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 17:45:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bcb76c193f946229202d8f9ba0c23c3 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6f91249b2ec468a805f12b03f3eed6e : SUCCESS in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71b47d9d01eb800f83baff83df89e40636feb4b2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-19 18:29:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786927/1/iscsi/ : SUCCESS in 53m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/786927/1/smb/ : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"69e3f34ddf43fae991bc8686b109aa01c7b51a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 19:26:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f8a495475ee4e1090194c6fbce20929 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dacdaa914fb4e3fb7128efd141ecc48 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b17bd4dd02d4e85aa73c52a20b56ce6 : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d922299ccc403ebe099635ce6e9b5f : SUCCESS in 11m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b11358647f94c6ca8669f13f1587664 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263358025d8d4b1cbb304b1310ae1796 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/341dbd23c05d413a976bfc47b85b1b83 : SUCCESS in 55m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eacfed801f9f4538bdee8b8b9f20f484 : SUCCESS in 1h 18m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a060ff202f22497986e89cb53d31f4d6 : SUCCESS in 16m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d18952732d1f4a8796ade6d366b70589 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20b8ee8bbcc74190a2c8afd7f0083f68 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a2fb5692b57f4fdeaf0343f5d4ce7d7d : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e6be80a0b3f40348a38ef97f4adc068 : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a10dc94bea4c4f08986d6a6ca43a7cb9 : FAILURE in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0225db0d7714bb39f49fcbc2138cfd1 : FAILURE in 1h 06m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/67653a2b22854975a79593f6c76c4842 : FAILURE in 1h 41m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6674befd22a40179001fe5246cbb1e6 : FAILURE in 1h 34m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/950d407b154840efbb6a0d251299812f : SUCCESS in 1h 26m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a1adabed3604bf2b46330c03735405f : SUCCESS in 1h 01m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2aa13cfd50b64c8aab36f1cc39af5222 : SUCCESS in 1h 03m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9cf1e742df1d46c380d5c0dccb030023 : SUCCESS in 1h 51m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/163f5c618ed54cfcb64822e3d1fd1b09 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2143a5daf2b68b4b7528726f25d8e9926c00fe81","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-19 20:50:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/27/786927/1/check/dsvm-tempest-cisco-zonemanager-job/19e7cdc/ : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5d03b9ae8fbefae47d8a26658ae2011617126372","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-19 21:29:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/27/786927/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7b700b5/ : FAILURE in 2h 14m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/27/786927/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c9dde9d/ : SUCCESS in 1h 58m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cb201ad3ae4697e39d407accc8c70e12447db9d4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-20 00:58:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-786927-1/ns5_nfs : SUCCESS in 4h 59m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-786927-1/ns5_iscsi : SUCCESS in 5h 1m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3d0e5996986856a1ca81f64b5e950ec7d6f229c8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 05:23:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/27/786927/1/check/DellEMC_XtremIO_FC/8533004/DellEMC_XtremIO_FC/10 : SUCCESS in 2h 04m 52s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/27/786927/1/check/DellEMC_XtremIO_iSCSI/77c5cb5/DellEMC_XtremIO_iSCSI/10 : SUCCESS in 1h 55m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"1333b76c64555bb08ee59f92d7ce40137c54e107","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 06:04:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/786927/1/check/iscsi-driver-nimble/fa563dd/fa563dd.html : FAILURE in 1h 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f98d799455a056c78b65b9639e38eb8b3537b351","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 07:32:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"2cac073f5030605eb2bdb038b422bafe0603c79e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 08:44:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/786927/1/20210420151614/NEC-ISCSI : SUCCESS in 2h 11m 50s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/786927/1/20210420151614/NEC-FC : SUCCESS in 2h 27m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b72773de6690f42b36151c42605ce5a435f3c20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-20 16:35:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e222e20ac937670f59f2945d12421b7075c07ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:33:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ebf1872308e4e6cae7aee3368abe771 : SUCCESS in 29m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/737f2f97bf744edda1ec6d398483ff5f : SUCCESS in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb5b7358653b0ba148e838f17b70201df8a0b654","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-20 17:48:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786927/2/iscsi/ : SUCCESS in 59m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/786927/2/smb/ : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":2},{"id":"6c8e2730e9cc3c7a73a81ae636459f84ff32c66f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-20 18:31:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/27/786927/2/check/dsvm-tempest-cisco-zonemanager-job/c04f4e8/ : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3101bf2a00cc1029071e17c64cd132ae4a4055c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 18:49:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/386c43fd9d6246dfba5a894091d2763e : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c593cad45ae42d98eab30dd6a1e3596 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2439ad4e3834a39ad9bebba11c13b14 : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6feb513aa5e74399b701c1befe2dcca2 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ecd6d483fbe460fab1207bf7885cd6e : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef566da7da5429eb630ed27fe7d6011 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/8d12e277e3c5419e86c2597b4ed19d01 : SUCCESS in 1h 06m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bca8308c3ba74f3397e7de635b877658 : SUCCESS in 1h 25m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8508481bf90b4ba08bd12ff2d870a779 : SUCCESS in 16m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3cf290d4706247e79038f196c99ffec9 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c133471399b4f51b6f69731dcad9bd6 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4987e18144a3430bbfade455564bb25d : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bbbe8a58ef2c443c98102de0789b7bdc : SUCCESS in 40m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98c5c74ff143413cb3215eed3ea3654e : FAILURE in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74be0a45a6de4c65b3149919aff8886e : SUCCESS in 1h 07m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9347c03f6a0c4fafb96c3d18e8ff606d : TIMED_OUT in 2h 04m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc388e8a98b341e8868203fc14bebd13 : SUCCESS in 1h 28m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db828469b6064d5cbb5e10782e646351 : SUCCESS in 1h 12m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/608863b81b4b43c7b7df4569a7f9d2e4 : SUCCESS in 1h 12m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e3d943cdca6b4da49ecb823b522daaac : SUCCESS in 1h 25m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b40b17484a64a2ab1436e818e104084 : SUCCESS in 1h 55m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa915ae88b3742ad8498ce59087dd736 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cadbadfc0f233f4c0e21692ec0587404321abfe3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 19:02:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/27/786927/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6ef60ac/6ef60ac.html : FAILURE in 55m 21s","accounts_in_message":[],"_revision_number":2},{"id":"36dc964de9032aaa8df6c137901b05c334556df0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-20 19:29:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/27/786927/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/937fea3/ : SUCCESS in 2h 20m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/27/786927/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d0895bb/ : SUCCESS in 2h 07m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1d7bb3ef497d2dfadc89fcd097ee0792383dab2a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-20 21:04:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/27/786927/2/upstream-solidfire-check/cinder-solidfire-iSCSI/0498161/ : FAILURE in 2h 12m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"dca2aaf04e32e9693e44ba33de941bbebb832b90","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-20 23:30:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/27/786927/2/check/iscsi-driver-primera/ddb5ed5/ddb5ed5.html : FAILURE in 1h 05m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4c0aa4fcdf03fa182f1c92cb666449b422181fcd","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-21 00:16:37.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786927-2/Lenovo-iSCSI : SUCCESS in 49:38.80\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786927-2/Lenovo-FC : SUCCESS in 45:24.91","accounts_in_message":[],"_revision_number":2},{"id":"67b88bee7ab362f4cc6ab258c0ae5473ff92d0b2","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-21 00:20:00.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786927-2/HPMSA-iSCSI : FAILURE in 1:05:40\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786927-2/HPMSA-FC : SUCCESS in 33:41.16","accounts_in_message":[],"_revision_number":2},{"id":"1db4928daa077cc51cd6297ac0e81a5196c06aa7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-21 00:57:13.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-786927-2/ns5_nfs : SUCCESS in 5h 10m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-786927-2/ns5_iscsi : SUCCESS in 5h 3m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f1b570a10f69fec8646cc79de37ed8a85d20595","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-21 01:03:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/27/786927/2/check/DellEMC_Unity_FC/9bd16c2/DellEMC_Unity_FC/14 : SUCCESS in 1h 17m 40s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/27/786927/2/check/DellEMC_Unity_iSCSI/912d0c4/DellEMC_Unity_iSCSI/14 : SUCCESS in 1h 34m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5adc5816dc42a0a6eacdf06084ef40df94a9b988","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-21 01:12:28.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786927-2/Seagate-iSCSI : SUCCESS in 46:21.99\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786927-2/Seagate-FC : SUCCESS in 40:03.78","accounts_in_message":[],"_revision_number":2},{"id":"c1e53de99093a2c9ee00dcec9d7e6f1d0d6d2e77","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-21 01:35:25.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786927-2/PVME-iSCSI : SUCCESS in 1:02:06\n* PVME-FC http://os-logs.tristero.net/dell-ci/786927-2/PVME-FC : SUCCESS in 51:56.75","accounts_in_message":[],"_revision_number":2},{"id":"1f1f4e609ed72d943605f115802ca8b35d739773","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-21 03:30:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/27/786927/2/check/DellEMC_SC_FC/864f973/DellEMC_SC_FC/14 : SUCCESS in 1h 59m 36s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/27/786927/2/check/DellEMC_SC_iSCSI/f6dcf9d/DellEMC_SC_iSCSI/13 : FAILURE in 2h 13m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"07988dfc7eb29e32380d5d58594cd2f0ea064823","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-21 03:49:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/786927/2/check/iscsi-driver-nimble/6328bdb/6328bdb.html : FAILURE in 58m 51s","accounts_in_message":[],"_revision_number":2},{"id":"54eda04bebcc2a37573993d1f35e2829f59ec016","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-21 06:46:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/27/786927/2/upstream-check/cinder-cDOT-iSCSI/c46f293/ : SUCCESS in 34m 01s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/27/786927/2/upstream-check/cinder-cDOT-NFS/2a25f28/ : SUCCESS in 33m 01s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/27/786927/2/upstream-check/cinder-cDOT-FCP/8ef88d2/ : SUCCESS in 55m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"915f3e637e15f435905ff0df06cbf9f7adfd1038","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 07:37:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/27/786927/2/check/DellEMC_PowerFlex/0d07fda/DellEMC_PowerFlex/13 : FAILURE in 2h 42m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9a99ed56f48dab1f128f8e96405d149e7e68ad7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 08:12:19.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/27/786927/2/check/tempest-dsvm-full-focal-py3/3f27699/ : SUCCESS in 1h 37m 36s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/27/786927/2/check/devstack-plugin-ceph-dsvm-focal-py3/d589dc1/ : SUCCESS in 2h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"438d2a2974b218888271e30062b4e375e76eca6a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 17:46:28.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/27/786927/2/check/DellEMC_VNX_iSCSI/cbf90eb/DellEMC_VNX_iSCSI/13 : SUCCESS in 2h 55m 52s\n- DellEMC_VNX_FC http://publiclogs.emc.com/27/786927/2/check/DellEMC_VNX_FC/acd5368/DellEMC_VNX_FC/15 : SUCCESS in 1h 31m 37s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"514c0aed67237817675fdfa709a3d8f28fcc9540","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 18:05:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/27/786927/2/check/DellEMC_XtremIO_FC/82c4879/DellEMC_XtremIO_FC/28 : SUCCESS in 2h 06m 02s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/27/786927/2/check/DellEMC_XtremIO_iSCSI/ea9c9c0/DellEMC_XtremIO_iSCSI/28 : SUCCESS in 1h 52m 53s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"6a709ed39aff8ac1e7ee45f89e9b0ab22f64671e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 18:36:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/786927/2/20210422011243/NEC-ISCSI : SUCCESS in 2h 13m 57s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/786927/2/20210422011243/NEC-FC : SUCCESS in 2h 23m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3161c80d84f5bb5449ea800ba2ff4139e38d9cb3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-22 11:52:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/27/786927/2/check/DellEMC_PowerMAX_FC/de9a6d3/DellEMC_PowerMAX_FC/10 : SUCCESS in 4h 21m 55s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/27/786927/2/check/DellEMC_PowerMAX_iSCSI/304232d/DellEMC_PowerMAX_iSCSI/10 : ABORTED in 10h 31m 26s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"dba037d1e93ab3ac8fb94a026a8d15a6f32e19df","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 20:58:46.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, that always bothered me a little. 😄","accounts_in_message":[],"_revision_number":2},{"id":"74eccd62fd6f120379b8ef2e1ea038bae41317e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-13 17:32:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9b9e99b381beac02c10e000119bd54a775b5cc12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 17:55:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5d2a202db54422bac725d51563f1973 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c1dd5deb6314720bd9feff9dd3120bc : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f184e3ee115f357d93b6c0e19a720a899ddd5df","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-13 18:29:43.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5891ef1b0ea340cf86f45e4d3674d287 : DISK_FULL in 50m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"c0d98e5df57399e88388698faac6b5919eea1ebf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-13 18:32:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786927/3/iscsi/ : SUCCESS in 57m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/786927/3/smb/ : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":3},{"id":"11db368ff3c99e9391133b5b0642f6d82e0d61ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-13 19:09:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/27/786927/3/check/iscsi-driver-primera/77ff4bd/77ff4bd.html : SUCCESS in 1h 21m 44s","accounts_in_message":[],"_revision_number":3},{"id":"82827e65fa1bfc2b672fb048e669a32ffefd5f40","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-13 19:16:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/27/786927/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4c7f3b5/ : SUCCESS in 1h 35m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/27/786927/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2ce78fa/ : SUCCESS in 1h 41m 59s","accounts_in_message":[],"_revision_number":3},{"id":"41235be04e5c7fcd677a9741039a50ea7befb345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 19:44:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f785fd5dbae4c81a7157a0e4c72d75e : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f17fde48ba34031b2c5a107c916a553 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ebb31bb986e4d74b6a1185a79d8bcbd : SUCCESS in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf5742a71ec945abbb38605b86769525 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b0c0b5a35c4bc89ff5f7688ab37bf3 : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/5d2aaeb804f24a19aee48ad4feba8326 : SUCCESS in 1h 05m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f782b303605b41db99b1bc2c4679fed6 : SUCCESS in 1h 05m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/048ca44697ad4f45a02d505f225353e4 : SUCCESS in 14m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2fe87d8bc45b41eea81fe5251eedf1df : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/718f9300174d43988b9e5592e66c2538 : SUCCESS in 7m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cdba773f8634426b7144318de78bb50 : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/49143c699eaa4fe6ab8cf9efdb820fa2 : FAILURE in 36m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cdfd5acc620e48bca19a176ac92faca4 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61f8001706ee4850b96ea3a8dd584cf0 : SUCCESS in 1h 31m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dae95561eb3d4c1787c154b597accc55 : SUCCESS in 1h 45m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbc061143b6a4c368b70cdc4abdd40aa : SUCCESS in 1h 19m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6b82f003bece49fe8c285752d8b1fa00 : SUCCESS in 1h 17m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c8526de1cd64b2a9ef8afdf1626f6ec : SUCCESS in 1h 09m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5ad63b4099254d078c1d614347ced9b8 : SUCCESS in 1h 15m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0fe41fd2b7d54174b678d8b8ca85d02f : FAILURE in 2h 06m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f38f8656d5914aff89b1b02090cfc583 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e65d573ea14593062b7f9fdd2c981751f5fac8e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-13 21:16:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/27/786927/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/dd7b747/dd7b747.html : SUCCESS in 2h 11m 18s","accounts_in_message":[],"_revision_number":3},{"id":"816487ec95fdc14195f29e7a498eeba9e8a527c5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-13 22:11:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/27/786927/3/ : SUCCESS in 1h 44m 37s","accounts_in_message":[],"_revision_number":3},{"id":"12ecd89755e82d619cc4a57d2e5515ae0705c345","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-14 00:21:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-786927-3/ns5_nfs : SUCCESS in 5h 9m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-786927-3/ns5_iscsi : SUCCESS in 5h 10m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"cd42908235490637eee40b152f2c2ab3b320d4cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 02:50:29.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"0dcef9a434d661763455b69a20e18d245140cf9d","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-07-14 03:51:10.000000000","message":"Patch Set 3:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":3},{"id":"2de3939bd38900d6b32d0333045ee8de85b701f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 05:09:40.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"3869f7438a4d26ed32d14926cb9bd26ed6de48fd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-14 05:48:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/27/786927/3/check/DellEMC_Unity_FC/20f42b2/DellEMC_Unity_FC/433 : SUCCESS in 1h 08m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/27/786927/3/check/DellEMC_Unity_iSCSI/bb1e8ae/DellEMC_Unity_iSCSI/408 : FAILURE in 1h 52m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8c6fce6b5103119fd4c95dc3b0f1f561ebf29204","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-14 07:47:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/27/786927/3/check/DellEMC_XtremIO_FC/b417ad1/DellEMC_XtremIO_FC/500 : FAILURE in 1h 19m 37s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/27/786927/3/check/DellEMC_XtremIO_iSCSI/d7c6c7b/DellEMC_XtremIO_iSCSI/490 : SUCCESS in 2h 00m 25s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"ea471f8dab071f3bebcee4d1ceb9f401430c8bef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-14 11:02:25.000000000","message":"Topic bp/alembic-migration removed","accounts_in_message":[],"_revision_number":3},{"id":"426f4ac31d55ccbf9549316747774e5758c70900","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-14 11:02:29.000000000","message":"Topic set to remove-sqlalchemy-migrate","accounts_in_message":[],"_revision_number":3},{"id":"b675f7e2aee764c51a6bb51daa13f6e9e7047dbf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-14 11:02:33.000000000","message":"Topic remove-sqlalchemy-migrate removed","accounts_in_message":[],"_revision_number":3},{"id":"e13eeaa470ef88b916b897385eacd96ef9000b77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-14 11:02:38.000000000","message":"Topic set to bp/remove-sqlalchemy-migrate","accounts_in_message":[],"_revision_number":3},{"id":"aeb3eeb9ce7cd2e4b6bad59af27da8706bf5c1c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-14 12:03:05.000000000","message":"Patch Set 3:\n\nrecheck tempest-slow-py3 - tempest.scenario.test_network_basic_ops.TestNetworkBasicOps.test_subnet_details","accounts_in_message":[],"_revision_number":3},{"id":"44f647addb57127aa62f0afae8b4643692602c01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 12:24:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79a1d1587e314f469797e54a161760b8 : SUCCESS in 16m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04510ab6af924e9f9ee33bf3434df1c9 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"efc3ccbfa030564011ac0dd34fdc1d85b0dc1cb5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-14 12:30:12.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"ae504ba7a9fd3a560e143dd54875ba8adad565f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 14:03:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d6e2563e32477fb2e81ad93872f710 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11cf1e8765d247ada54be1dcffc5e182 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3818e8ea64954dd2947899cb414faff4 : SUCCESS in 9m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eae637d9f5b4b72a03f0649606f1c26 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f203bcfdf54f81a662d52984dac0b5 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/d279b994caa2454da9580a0c41656a5b : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58876767e83a4ed0bb7125a085e72f26 : SUCCESS in 1h 12m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/926d6b71f0454cbd8e0a9e52508bb1b1 : SUCCESS in 11m 32s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e468675f731d4ab3bba96a09d0d12ada : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/416dac7e04d34062a1322595385213bc : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aefdcc44f15a4e4b8ea604993a29f610 : SUCCESS in 7m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1224d28f5ff840cb85a7431f0caf5348 : FAILURE in 34m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/984104f624834ec286b6ce0772e6a0f0 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd3fb74b1229460998230f593312a8cc : SUCCESS in 1h 05m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80d9c5a25ca24b1682a99787827c2082 : SUCCESS in 1h 32m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/deed2ad7c3e34780a121efdca405e87b : SUCCESS in 1h 37m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c22a9c6c1944f2893c0c9bcd106612d : SUCCESS in 1h 05m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88db9005b9db4da4a4f415beb60cc582 : SUCCESS in 1h 08m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e764d8b59b04fa0a9910ccad323887b : SUCCESS in 1h 09m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ffaa81b46f94516859a0a20f7242a50 : SUCCESS in 1h 59m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9544b9a7820450686458df1074a2a72 : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ffe3732a018b421ababd19fcfc41f2b83a0daef7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-14 14:29:08.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/786927/786927/3/check-not-vote/ext-cinder-zuul/568a036 : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":3},{"id":"cf1df518498e222f2da8789384282e9b009762ea","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-14 14:44:16.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/786927/3/iscsi/ : FAILURE in 1h 20m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/786927/3/smb/ : SUCCESS in 49m 41s","accounts_in_message":[],"_revision_number":3},{"id":"8c0983f9f6be9cb3be042e0d2881a714187664b4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-14 15:32:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/27/786927/3/upstream-check/cinder-cDOT-iSCSI/23ff75c/ : SUCCESS in 34m 15s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/27/786927/3/upstream-check/cinder-cDOT-NFS/189bdc3/ : SUCCESS in 39m 06s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/27/786927/3/upstream-check/cinder-cDOT-FCP/0e3ccf9/ : SUCCESS in 1h 28m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"fd885e2a6562fb30d0a5c1b7fe2252196d3b704d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-14 15:49:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/27/786927/3/upstream-solidfire-check/cinder-solidfire-iSCSI/5c917d8/ : FAILURE in 1h 45m 13s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"1af3b212e9f6757fd33ca4fe906e865da8e9a2b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 17:24:34.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b867c87173545769f9110f523c1eb30a837dcaa6","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-14 22:11:50.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/786927-3/PVME-iSCSI : FAILURE in 54:59.37\n* PVME-FC http://os-logs.tristero.net/dell-ci/786927-3/PVME-FC : SUCCESS in 47:41.48","accounts_in_message":[],"_revision_number":3},{"id":"4a053c00f48574516ef8e643bdd3b115d4fee0ea","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-14 23:58:10.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/786927-3/Seagate-iSCSI : SUCCESS in 41:33.76\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/786927-3/Seagate-FC : SUCCESS in 39:41.05","accounts_in_message":[],"_revision_number":3},{"id":"062c3c348e08a8efc889f5f053d03adaafa0b3fb","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-15 00:58:02.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/786927-3/HPMSA-iSCSI : FAILURE in 1:00:52\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/786927-3/HPMSA-FC : SUCCESS in 31:45.57","accounts_in_message":[],"_revision_number":3},{"id":"dde1da1d463c9997760e38b09fcf0acafebec996","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-15 12:13:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/27/786927/3/check/DellEMC_Unity_FC/dad7581/DellEMC_Unity_FC/461 : SUCCESS in 1h 08m 43s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/27/786927/3/check/DellEMC_Unity_iSCSI/e73eba7/DellEMC_Unity_iSCSI/432 : FAILURE in 1h 16m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5bc772b8f10cea9a2cfec100118d99e1e807217d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 13:35:35.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c5053e9437e23819f1647727d543ccb5da4dfe99","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 15:58:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/27/786927/3/check/DellEMC_XtremIO_FC/8c8f52c/DellEMC_XtremIO_FC/525 : FAILURE in 1h 17m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/27/786927/3/check/DellEMC_XtremIO_iSCSI/92ef40a/DellEMC_XtremIO_iSCSI/511 : FAILURE in 7m 54s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"bdd7116ae1826d9716a5b2e347c475c5201c3dea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 17:24:53.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"62672a624d6a740251f115a032ba8ed0ac41d14d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-16 02:36:49.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/786927-3/Lenovo-iSCSI : SUCCESS in 52:18.53\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/786927-3/Lenovo-FC : SUCCESS in 50:07.79","accounts_in_message":[],"_revision_number":3},{"id":"3c729f1479e356b05ee5056e0309e215a735c7bc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 03:53:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/27/786927/3/check/DellEMC_PowerMAX_FC/728531d/DellEMC_PowerMAX_FC/259 : FAILURE in 5h 02m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/27/786927/3/check/DellEMC_PowerMAX_iSCSI/6e41613/DellEMC_PowerMAX_iSCSI/243 : FAILURE in 2m 10s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"49942e84b9f27b49f537fd8949a85243c45ee887","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 05:25:28.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/27/786927/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"27b5c32ca3679a20b31100c8f2f9ee0f0caa3cbe","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-16 07:03:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6ffa8c0d026e58f1a95bca307d3bb22218ffd2d6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 23:09:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/786927/3/20210717032700/NEC-ISCSI : SUCCESS in 2h 40m 10s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/786927/3/20210717032700/NEC-FC : SUCCESS in 4h 42m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2f5bbd4b8edb5573ea5a4258ce0092d05c804762","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-04 15:11:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2b75f6d80a13f6b76c864386746419a2c5a5cdae","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-25 09:40:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nRefactoring and not changing any functionality. LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"51d7aed1a1804537f56da55f7730c91d543c127e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:41:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c80f9c1f0261039f0f294a4c73b8b1f95510409","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:04:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd20b5097a047f48e8a60d5722c8ee7 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1ce962139be4b4b9dc9cd44642e89ea : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def45ac1ac5a4fa4a34b002b211d30f7 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e538b03f89c4f51bfb6f8812bd44d3b : SUCCESS in 15m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/b6bf0000575145b788a8a7e114eed39a : SUCCESS in 1h 05m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcfb7cdc958b4a80ac86dd8c13834be2 : SUCCESS in 1h 08m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4f90930d7eee4736a87f50ec04aa3d7e : SUCCESS in 1h 12m 28s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6af478432ac44928bce461f67d40790b : SUCCESS in 1h 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dec8f920571c48189058902fc1d15ff0 : SUCCESS in 52m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1028510140544728b7ec3603229bec2 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b879c16ec33fb22c973588c8c634ff3bf0ec7733","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:04:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8d1bc34f342273c400e6bb043d7995ba11e45ca8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:06:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f47cf63876474daec0e3f6ee1b3c75 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b2993cdddc6831baefe498e207f28c62cff2ee1","revisions":{"a6111f5a1165b86f90f2c863b88ceee9f6f18034":{"kind":"REWORK","_number":1,"created":"2021-04-19 16:57:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/786927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/786927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/786927/1"}}},"commit":{"parents":[{"commit":"8e6b86b7c902a27e855708ffe2ffc4cc638dcef9","subject":"db: Compact migrations to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e6b86b7c902a27e855708ffe2ffc4cc638dcef9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 11:08:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 14:22:11.000000000","tz":60},"subject":"db: Use \u0027import sqlalchemy as sa\u0027 pattern","message":"db: Use \u0027import sqlalchemy as sa\u0027 pattern\n\nThis makes it a little clearer where things are coming from.\n\nChange-Id: Icb1cf73722e0e9bf1364d64522e38b94aa716bea\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6111f5a1165b86f90f2c863b88ceee9f6f18034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6111f5a1165b86f90f2c863b88ceee9f6f18034"}]},"branch":"refs/heads/master"},"a8449535c618fdb1b408cc35a0cf9135de51cc8f":{"kind":"REWORK","_number":2,"created":"2021-04-20 16:35:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/786927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/786927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/786927/2"}}},"commit":{"parents":[{"commit":"b29bb44e98e00c68382b3f381a0a6e0938fe6dda","subject":"db: Compact migrations to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b29bb44e98e00c68382b3f381a0a6e0938fe6dda"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 11:08:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-20 16:13:22.000000000","tz":60},"subject":"db: Use \u0027import sqlalchemy as sa\u0027 pattern","message":"db: Use \u0027import sqlalchemy as sa\u0027 pattern\n\nThis makes it a little clearer where things are coming from.\n\nChange-Id: Icb1cf73722e0e9bf1364d64522e38b94aa716bea\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8449535c618fdb1b408cc35a0cf9135de51cc8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8449535c618fdb1b408cc35a0cf9135de51cc8f"}]},"branch":"refs/heads/master"},"2b2993cdddc6831baefe498e207f28c62cff2ee1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-13 17:32:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/786927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/786927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/786927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/786927/3"}}},"commit":{"parents":[{"commit":"1b5aad85d04b262281b9632bbc9a3ab490ee9082","subject":"Merge \"Abort volume creation when encryption spec is invalid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b5aad85d04b262281b9632bbc9a3ab490ee9082"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-19 11:08:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-12 15:16:55.000000000","tz":60},"subject":"db: Use \u0027import sqlalchemy as sa\u0027 pattern","message":"db: Use \u0027import sqlalchemy as sa\u0027 pattern\n\nThis makes it a little clearer where things are coming from.\n\nChange-Id: Icb1cf73722e0e9bf1364d64522e38b94aa716bea\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b2993cdddc6831baefe498e207f28c62cff2ee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b2993cdddc6831baefe498e207f28c62cff2ee1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
