)]}'
{"id":"openstack%2Fcinder~788033","triplet_id":"openstack%2Fcinder~master~I6befa027626dc468eeda8a8f10af0be10caf661a","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"I6befa027626dc468eeda8a8f10af0be10caf661a","subject":"filter_scheduler: Remove schedule() method","status":"MERGED","created":"2021-04-26 15:35:38.000000000","updated":"2021-05-01 06:46:29.000000000","submitted":"2021-05-01 06:43:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788033","meta_rev_id":"102dbf7750ae868d86980da53b8b21eb46f5ed60","_number":788033,"virtual_id_number":788033,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-01 06:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-04-30 16:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-30 15:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-04-30 16:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 15:49:38.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 15:59:03.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2021-04-26 16:01:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 16:20:04.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 16:54:13.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-26 17:17:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-26 17:23:17.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-26 17:25:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 18:03:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 18:04:40.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-26 18:40:23.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 18:49:34.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-04-26 20:02:43.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-26 20:10:21.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-26 20:11:25.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-26 20:11:37.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-26 20:28:53.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-26 20:57:04.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 21:05:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 22:20:13.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 23:18:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 22:28:08.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 15:21:50.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-04-30 16:20:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"98b83a40e6125fa4a30d960afa12e4332e88a0a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-26 15:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2539b61e99a67a697c336b5334e8485593cb63d6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-26 15:49:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/33/788033/1/check/cinder-storpool-tempest/3b1b107/ : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bb87ddbf81569840376ed949c31355ccc0b0f1a3","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2021-04-26 15:59:03.000000000","message":"Patch Set 1:\n\nrun-storpoolci","accounts_in_message":[],"_revision_number":1},{"id":"a8257bd9bcab2a2e76c265195ddec5743426bb18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:01:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d73a590f74c4a1b941bb72be5e07928 : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e4dc3214b154ee0b94d28b24c35cd65 : SUCCESS in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cf3a051823b9ecdcf0efc47f9931a786129fd6b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-26 16:20:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/33/788033/1/check/iscsi-driver-nimble/48b5b3d/48b5b3d.html : FAILURE in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"74deb1fe1b38a255a310f47fc625ea235e06a136","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-26 16:54:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/33/788033/1/check/dsvm-tempest-cisco-zonemanager-job/cc02694/ : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"01bab0a2d8667aca4b58a58d1ea267fc27e7639c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-26 17:17:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/33/788033/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bfa25e3/ : SUCCESS in 1h 40m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/33/788033/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e84741d/ : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1442253025d8ce327d266f70b775a4a67b557dbf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-26 17:23:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/33/788033/1/check/DellEMC_Unity_FC/e811d74/DellEMC_Unity_FC/51 : SUCCESS in 1h 23m 18s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/33/788033/1/check/DellEMC_Unity_iSCSI/697b026/DellEMC_Unity_iSCSI/50 : SUCCESS in 1h 45m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"277cd2ceb4ae2765235e4b754540fc8d028c2787","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-26 17:25:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/33/788033/1/check/tempest-dsvm-full-focal-py3/f9a790a/ : SUCCESS in 1h 31m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/33/788033/1/check/devstack-plugin-ceph-dsvm-focal-py3/85a650e/ : SUCCESS in 1h 38m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1191a7454085c9fcc449409a02272ee4db06d0c9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-26 17:28:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/33/788033/1/check/cinder-storpool-tempest/2f5e70d/ : POST_FAILURE in 1h 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fa4d56a9bfe99f227f16a16bb56ead5b09501ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:03:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b2dd113fbb4961be58e73e25fa2190 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac56894bc01f4458997aae4891c120f8 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/903c1ade4a844fcf999336ff4cd834b9 : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e3a64c9966a452490e7e4dee61897ce : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ba55ac0710f4b8cb592beeee27de583 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b209c2eecb47f78194b519d64214a1 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/56de2db4dda54fe89a7bc6805952bf6b : SUCCESS in 1h 03m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a9f7d08a0294019a0565f668ca06867 : SUCCESS in 1h 15m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/58ab51246d5f4ca3885529ca8c928d84 : SUCCESS in 12m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14c18f58bc92468c9fbc3a87430b4a38 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91074561bcbd4d3eaf4de22f755d5a41 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6fd72970d9749fba7c20213d6f05d5f : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c2971fc5361c4033914c0962fbe0c306 : SUCCESS in 35m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2abea69588a4d8e946df16e761f6992 : SUCCESS in 4m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ba2f85d6c0743b6b9397f4113204cf9 : SUCCESS in 1h 24m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d8378efaed954d74b0801ed87f26f3d6 : POST_FAILURE in 20m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3c5745f2b12c43a1bfec3eef8ac43cde : SUCCESS in 1h 28m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c4ae9529c9b4db9b4f434492179b513 : SUCCESS in 1h 14m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8ba1c2b6c6f4fd8bc09e3f05d9e5438 : SUCCESS in 1h 20m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bfa318b2f7d34bd698da2b808d3603d8 : SUCCESS in 1h 26m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac80e92c80604defa7eb275cc0c2804d : SUCCESS in 2h 02m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66762614f6934dc78ca7841544234f13 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":1},{"id":"42e67bc4f37ded388b1699847d1c9c0e6c115815","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-26 18:04:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/33/788033/1/upstream-solidfire-check/cinder-solidfire-iSCSI/56d23b9/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1da6039a0551cbf44eeb98ac223a843cc7c94db","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-26 18:40:23.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/33/788033/1/check/DellEMC_XtremIO_FC/f8bd80c/DellEMC_XtremIO_FC/61 : SUCCESS in 2h 07m 40s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/33/788033/1/check/DellEMC_XtremIO_iSCSI/d6749bf/DellEMC_XtremIO_iSCSI/61 : SUCCESS in 1h 55m 52s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"b62a3b8ef8db678b4480c50f058e40e5e1ae5d56","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-26 18:49:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/33/788033/1/check/DellEMC_PowerStore_iSCSI/44955f6/DellEMC_PowerStore_iSCSI/5 : SUCCESS in 44m 30s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/33/788033/1/check/DellEMC_PowerStore_FC/a20dc10/DellEMC_PowerStore_FC/6 : SUCCESS in 45m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c5784d74d9a9bdd46cfb7e614e1822dd35276c7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-26 19:45:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/33/788033/1/check/DellEMC_Unity_FC/075b963/DellEMC_Unity_FC/52 : SUCCESS in 1h 41m 45s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/33/788033/1/check/DellEMC_Unity_iSCSI/8736e5c/DellEMC_Unity_iSCSI/51 : SUCCESS in 1h 34m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ada9820378d85293cdcc529e91cf3a5d6462ef7","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-26 20:02:43.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788033-1/Seagate-iSCSI : SUCCESS in 47:06.48\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788033-1/Seagate-FC : SUCCESS in 40:39.87","accounts_in_message":[],"_revision_number":1},{"id":"e05e425fc5026f574dee0f4ddb14c7de4679e6a4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-26 20:10:21.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788033-1/HPMSA-iSCSI : FAILURE in 1:01:49\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788033-1/HPMSA-FC : SUCCESS in 34:19.44","accounts_in_message":[],"_revision_number":1},{"id":"a6fcecc87441911477684e171cf1e5ec29a5bf8e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-26 20:11:25.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/33/788033/1/check/DellEMC_SC_FC/9ce2096/DellEMC_SC_FC/37 : SUCCESS in 2h 06m 51s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/33/788033/1/check/DellEMC_SC_iSCSI/d86f493/DellEMC_SC_iSCSI/34 : SUCCESS in 2h 02m 21s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c214334478c8acdde799891919d766d13d065f7","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-26 20:11:37.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788033-1/Lenovo-iSCSI : SUCCESS in 49:54.79\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788033-1/Lenovo-FC : SUCCESS in 45:50.91","accounts_in_message":[],"_revision_number":1},{"id":"d9738f0f92aba16d58b7507dd73307eba73209f1","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-26 20:28:53.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788033-1/PVME-iSCSI : SUCCESS in 1:01:18\n* PVME-FC http://os-logs.tristero.net/dell-ci/788033-1/PVME-FC : SUCCESS in 49:31.60","accounts_in_message":[],"_revision_number":1},{"id":"fb57f4189f03523867db1e520e7c7b0d7bdd36b9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-26 20:57:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/33/788033/1/check/DellEMC_PowerFlex/bbb411e/DellEMC_PowerFlex/18 : FAILURE in 2h 34m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c2abf6a99b19ae4aeb2501836f3943f80796f7e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-26 21:05:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/33/788033/1/20210427034526/NEC-ISCSI : SUCCESS in 2h 18m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/33/788033/1/20210427034526/NEC-FC : SUCCESS in 2h 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1992fcf9812a00fb79aee5dce7638a82f5a0d13a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-26 22:20:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/33/788033/1/check/DellEMC_VNX_iSCSI/6cc48c7/DellEMC_VNX_iSCSI/33 : SUCCESS in 3h 05m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/33/788033/1/check/DellEMC_VNX_FC/03c0a2f/DellEMC_VNX_FC/36 : SUCCESS in 1h 42m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"a90fb11e40df03bcd5152d6ebd0ac8923342d2bb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-26 23:18:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-33-788033-1/ns5_nfs : SUCCESS in 4h 46m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-33-788033-1/ns5_iscsi : SUCCESS in 4h 40m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4608913fb364c0d35fb11bd6013aa96fbbb571e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-04-28 22:28:08.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788033/788033/1/check-not-vote/ext-cinder-zuul/41a2e97 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0cb7e726d3b3b751e8cfecfdc1591bee26d65e0f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-30 15:21:50.000000000","message":"Patch Set 1: Code-Review+2\n\nClean-up is good.","accounts_in_message":[],"_revision_number":1},{"id":"2857e8f4cd3557a63ef84f9e2052ffdb5d1ce75b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-30 16:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good.","accounts_in_message":[],"_revision_number":1},{"id":"f4de0a599100c319bcdfbece1094cdaea3ff8c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 16:44:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e982cf74ffe054cab5361f300dece07741ffb621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 20:19:18.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"188d079db1968c9d7edb4ff8b38d17b25c957622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 06:43:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3d25fd8b681499fa9955420db289949 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d0f8e62168461ab2e2e3205922e504 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9da2b4545775432098a1015b748b59a5 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b011b2b5fcf445591b48e5f1217d4c9 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cc865b643e46ada126b665e38dea56 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/59101eafbff347ec93b4db47110da564 : SUCCESS in 1h 02m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff26d8bd5fc945de86075933bd4e6fe3 : SUCCESS in 1h 08m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/10d24f0b77cd4b1cb1a41019ad4ffef1 : SUCCESS in 1h 03m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8225eeb7322b4642b25b58f70c892330 : SUCCESS in 1h 14m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5eafe399a261447d9247d19abd2252e4 : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8bc5d0e8d07a8ca320b4762e8df63d76cf935395","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 06:43:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"102dbf7750ae868d86980da53b8b21eb46f5ed60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 06:46:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2852d8cd1134859a0fbb4fdcc159a67 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b64fef783f808c5a1611a03b77742906d0df6713","revisions":{"b64fef783f808c5a1611a03b77742906d0df6713":{"kind":"REWORK","_number":1,"created":"2021-04-26 15:35:38.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/33/788033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/788033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/788033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/788033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/788033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/788033/1"}}},"commit":{"parents":[{"commit":"ea7ae8e65598290c317717f290c9db09edd9daf4","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea7ae8e65598290c317717f290c9db09edd9daf4"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-26 15:20:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-26 15:31:51.000000000","tz":-240},"subject":"filter_scheduler: Remove schedule() method","message":"filter_scheduler: Remove schedule() method\n\nThis is never actually called, only _schedule()\nor schedule_create_volume() are used.\n\nIt also doesn\u0027t appear to work since it passes\nthe topic as the second parameter to _schedule(),\nwhich should be a request_spec.\n\nChange-Id: I6befa027626dc468eeda8a8f10af0be10caf661a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b64fef783f808c5a1611a03b77742906d0df6713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b64fef783f808c5a1611a03b77742906d0df6713"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
