)]}'
{"id":"openstack%2Fcinder~788363","triplet_id":"openstack%2Fcinder~master~Iebb002cba51fc0f5e42565151a1687da20bd5a24","project":"openstack/cinder","branch":"master","topic":"mypy","hashtags":[],"change_id":"Iebb002cba51fc0f5e42565151a1687da20bd5a24","subject":"mypy: image cache","status":"MERGED","created":"2021-04-27 21:46:11.000000000","updated":"2021-08-21 08:31:42.000000000","submitted":"2021-08-18 22:29:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"788363-mypy","meta_rev_id":"15021d27be12f02c0a9ef3229f1849bdccfa120b","_number":788363,"virtual_id_number":788363,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-18 22:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-08-10 21:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-10 22:39:36.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 22:39:36.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 22:08:23.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 22:08:23.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-27 22:09:25.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-04-27 22:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 23:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 23:42:37.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-04-28 00:02:29.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-04-28 00:03:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 00:31:06.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-04-28 00:45:28.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 00:53:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 01:24:46.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-28 01:33:01.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-04-28 01:34:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 01:44:30.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-04-28 01:45:56.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-04-28 02:06:34.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-04-28 02:10:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 02:23:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-28 06:58:51.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 14:08:01.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 13:32:28.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-04-30 14:31:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 15:04:59.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-04-30 23:42:38.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 15:54:51.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-06-23 17:03:05.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-23 23:24:45.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-06-24 00:47:25.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2021-06-24 07:39:26.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 21:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-07-12 16:44:29.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-09 22:28:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-10 06:58:40.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"}],"messages":[{"id":"e8831cb525fba28a5d4835df2123cd0733f2236a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-27 21:46:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"245186d4efc2ea83861e7944e9d97960544b09e9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-27 22:08:23.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"40b08e1b6cee6e1739c0cf209d99fbaaf89ad072","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-27 22:08:23.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"536687917b5745fa9fb1247ba40c380a457660ff","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-04-27 22:09:25.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a90b542fba79962929a816e0dcd29c28121037c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 22:17:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29db3692d79541fca4253668032ae21a : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb3bf99d7fc6447388bd19013c8e5c7e : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3e9731ee1e1ed4499f143cf674da4185d235f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 23:39:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b21bf40917dd484f839f84f3449e776e : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71714cf17604638a2a52cc5774f9cb1 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5276a1e2ecd45e5a4e62b3ebb749e68 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2a4a5773ff84afaa7e8976add24187f : SUCCESS in 13m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/479c253c29764c299fdd686e2a007801 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545a59b5367248d1a45b88325c73b2f0 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/e76365f367ac45f1aa98df857f0ea945 : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c1d739924984228911aa850b549f91f : SUCCESS in 1h 02m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59a888824eab40fe9c36a5f745292dba : SUCCESS in 15m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/70ff3a43800848bdbdd1ffdde3b58aae : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b33df2795cb477d90b0fa97eb907272 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c25cc58dc584de5a20323f09b65577a : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/206b087c1c9c4e0eaf4c26958d224369 : SUCCESS in 39m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/039cd1f04bf048b09dd732b0b7a028e7 : SUCCESS in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d8e078ae8e34e84b88ceb7aababa1f8 : SUCCESS in 1h 11m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11d9106c295c49adb34d34563a9178b7 : POST_FAILURE in 18m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f08bf5f3234b47a9828b613f3acb5814 : SUCCESS in 1h 25m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d7de3ac2528d4828b7bdb21572d24d40 : SUCCESS in 1h 12m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ec6fc3dad3204f2fbe17b0ed8385229e : SUCCESS in 1h 09m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/465d091b251f4a96ba444d181ab7accd : SUCCESS in 1h 15m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e70faf9537e74073b42e97098c9d1c51 : SUCCESS in 1h 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7be2c59ced4ec1a07dab86fd414145 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"30efc857d6a49730c8285e85ea32e87983a873af","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-27 23:42:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/1/upstream-solidfire-check/cinder-solidfire-iSCSI/c9c91ac/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"78895ee7b6582a46dd5cad65df4758f1d0add5db","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-28 00:02:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/63/788363/1/check/DellEMC_SC_FC/313b535/DellEMC_SC_FC/42 : FAILURE in 22m 34s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/63/788363/1/check/DellEMC_SC_iSCSI/243a204/DellEMC_SC_iSCSI/39 : FAILURE in 20m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"52b1082a35db75e2ed47ddb8a4384e6d4bdab739","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 00:03:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/63/788363/1/check/DellEMC_PowerFlex/4e60779/DellEMC_PowerFlex/23 : FAILURE in 23m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1a8efa8b4422d00728e3c83bec8e78e56d791d2a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-28 00:31:06.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/1/check/DellEMC_PowerStore_iSCSI/717eec9/DellEMC_PowerStore_iSCSI/10 : SUCCESS in 46m 23s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/1/check/DellEMC_PowerStore_FC/24317bf/DellEMC_PowerStore_FC/11 : SUCCESS in 50m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"61ced4a6c5946e5a9b55a99e2ddef1fb89ab5fa2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 00:45:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/63/788363/1/check/DellEMC_VNX_iSCSI/d34e41d/DellEMC_VNX_iSCSI/38 : FAILURE in 1h 05m 35s\n- DellEMC_VNX_FC http://publiclogs.emc.com/63/788363/1/check/DellEMC_VNX_FC/d8a740a/DellEMC_VNX_FC/41 : FAILURE in 24m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d32043a2aa1a741309f8c89f6c57dfb910f4303","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-28 00:53:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/1/check/cinder-storpool-tempest/e278103/ : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6e71c5872672d57071e62bb4458bdd7325084de6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-28 01:24:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/1/check/dsvm-tempest-cisco-zonemanager-job/576e6be/ : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a364963596c52151de06969e1ac96675a678d015","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-28 01:33:01.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-1/Seagate-iSCSI : SUCCESS in 45:32.29\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-1/Seagate-FC : SUCCESS in 39:46.47","accounts_in_message":[],"_revision_number":1},{"id":"6bd6d357d4bfa41d0214e90817eb8834b10998fc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-28 01:34:54.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/1/check/tempest-dsvm-full-focal-py3/4ed6e0c/ : SUCCESS in 1h 39m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/1/check/devstack-plugin-ceph-dsvm-focal-py3/de98002/ : TIMED_OUT in 3h 08m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9654f9f58169310f9a7d66ad40972c3590482b9e","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-28 01:44:30.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-1/Lenovo-iSCSI : SUCCESS in 50:50.50\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-1/Lenovo-FC : SUCCESS in 46:29.64","accounts_in_message":[],"_revision_number":1},{"id":"97b7835818db985aafd5a54a7513b678dfba08ec","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-28 01:45:56.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-1/HPMSA-iSCSI : FAILURE in 1:03:41\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-1/HPMSA-FC : SUCCESS in 34:33.13","accounts_in_message":[],"_revision_number":1},{"id":"2d8f955f92c67b9f2cc3492c6bccdcf000d966e2","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-28 02:06:34.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-1/PVME-iSCSI : SUCCESS in 1:01:51\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-1/PVME-FC : SUCCESS in 56:07.55","accounts_in_message":[],"_revision_number":1},{"id":"1dea4f494ad0f4e0592ed23f4e8528b952e5cfcc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-28 02:10:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/63/788363/1/check/iscsi-driver-nimble/ba90955/ba90955.html : FAILURE in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"54e3a94c173f8614644e530290aacc815473c129","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-28 02:23:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/96d474b/ : SUCCESS in 2h 08m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a678899/ : SUCCESS in 2h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"095cd96351ff841f6bb808329b555716bedeba2b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-28 06:58:51.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-1/ns5_nfs : FAILURE in 53m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-1/ns5_iscsi : SUCCESS in 4h 57m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c63c2f74e7f3a5323388051725ab51301f622b7f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-28 07:53:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/1/check/DellEMC_Unity_FC/a7330cc/DellEMC_Unity_FC/64 : SUCCESS in 1h 20m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/1/check/DellEMC_Unity_iSCSI/77c825c/DellEMC_Unity_iSCSI/63 : SUCCESS in 1h 45m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee679f45bde4365a11e0204b4ad08f102ad9b165","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 08:14:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/1/check/DellEMC_XtremIO_FC/27476ec/DellEMC_XtremIO_FC/73 : FAILURE in 2h 07m 34s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/1/check/DellEMC_XtremIO_iSCSI/26fd639/DellEMC_XtremIO_iSCSI/73 : FAILURE in 6m 26s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"da5c22f1bd5d9e15a2f509331bfce2bd5a59d7ea","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 14:08:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/1/20210428203730/NEC-FC : SUCCESS in 2h 30m 06s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/1/20210428203730/NEC-ISCSI : SUCCESS in 2h 30m 29s","accounts_in_message":[],"_revision_number":1},{"id":"050d71f5c1e5a8e5997a7d068e7ffcc72c2f7493","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-28 23:19:34.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-1/ns5_nfs : FAILURE in 5h 45m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-1/ns5_iscsi : SUCCESS in 5h 37m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"03b838df5b7e5efac74cf4dcd0acf04e306744d0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-29 05:11:40.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-1/ns5_nfs : SUCCESS in 4h 52m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-1/ns5_iscsi : SUCCESS in 4h 38m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4e4391506c4cf935c81de97d8d14a99ca0d9ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-30 12:26:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fcc4eddc04e02001009cac4731e499d884a744ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 12:52:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/369b442ec7054888ab96c040f40ef1d3 : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/feb83e277de3469a834a3f1f090c0632 : SUCCESS in 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3506461c70d48c60b3724b5b3f4f629583c14991","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-30 13:32:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/2/iscsi/ : SUCCESS in 1h 02m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/2/smb/ : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":2},{"id":"77222b8f93a8070c474a7fdad5527eb89f5e8243","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 14:24:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/08f5be7/08f5be7.html : FAILURE in 55m 24s","accounts_in_message":[],"_revision_number":2},{"id":"729f4b704cdaa77ae415d9411606384686cc279d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:29:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb99cc76b8a4453aa30b360a871d8374 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d73317b9af42218df995133067bbb8 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9eb14923b484f4b8695cb6fe91cc74b : SUCCESS in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec542c376dfe46f7a40ed65722f6cf7f : SUCCESS in 9m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acb45d6272f64a4094240b896c241ea1 : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616b6e472ff64ab3bc3a617974b73d98 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/a51c410c0ce8454786fb3bd5771dc913 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1e236b945ed4d0e8a26284be04eb4db : SUCCESS in 1h 20m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/82fe3833a0fc47158e15220b81beada0 : SUCCESS in 13m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f06d1888f9114f0f9c5f4ec85c3dd312 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2276cdfde8c64eefb4b9eeecc799c6d4 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98fcaf5c6aa740a48fbe8ac40254bd8e : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a9c5b3f0300c4f17ace047545476733f : SUCCESS in 36m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11831ec2859d49fcbe167cfd8f7c143a : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68d47aa311c14ad99d812535ca278204 : SUCCESS in 1h 17m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/135a3f8d365b4c248125d56edddcb4f4 : FAILURE in 1h 27m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10c6bef0529e4202a19f61e9a59518c9 : SUCCESS in 1h 22m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1527c4907ffb45458787ee0a597eccab : SUCCESS in 1h 22m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09f0087fd84344cfa902438446dbefb8 : SUCCESS in 1h 04m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e0de731e4b11494cbaea958fab9a15c3 : SUCCESS in 1h 16m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8fe2fce5935e46c3ba7ab5f0dd4d181a : SUCCESS in 1h 57m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f161f68b6ad4d85bd02c690731676ca : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":2},{"id":"abaeb22bbe8f8da27cb522f0b22c4b22ad39a7b1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-30 14:31:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/2/upstream-check/cinder-cDOT-iSCSI/a73e1e2/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/2/upstream-check/cinder-cDOT-NFS/b82f03b/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/2/upstream-check/cinder-cDOT-FCP/c400391/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"67f83759658b6a549f78f012c5e96a8bfb3c2a2c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-30 14:32:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/2/upstream-solidfire-check/cinder-solidfire-iSCSI/a42fd39/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"173114660709ea4769238cb5d940fb2b563d1422","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 14:37:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/2/check/iscsi-driver-primera/80fa9ed/80fa9ed.html : FAILURE in 1h 02m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f64ab30e5802485d31c8cd029ce658bbb7e5279a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-30 14:51:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f1dea4f/ : POST_FAILURE in 2h 20m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a6c8e74/ : SUCCESS in 2h 08m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2743e7342fb42d44726f74ca634c0947271a4426","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-04-30 15:04:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nLooks good.  One minor thing, just missing a return type.","accounts_in_message":[],"_revision_number":2},{"id":"bdaa54cd2c383521303acdf528748feadd0748fa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-30 15:09:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/2/check/dsvm-tempest-cisco-zonemanager-job/ba128c3/ : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":2},{"id":"079929e98674afae99d16f35fba1a0a8d9a42a01","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-04-30 15:16:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/2/check/DellEMC_PowerStore_iSCSI/7ac391d/DellEMC_PowerStore_iSCSI/28 : SUCCESS in 46m 23s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/2/check/DellEMC_PowerStore_FC/a3dd477/DellEMC_PowerStore_FC/29 : SUCCESS in 46m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"46eef598949b2992163571eab494a12b78020aa8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-30 15:17:41.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7aceaa28c24a4f8eb7b243628a40a282 : RETRY_LIMIT in 29m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f35dda3be62d45248570768acab6089c : POST_FAILURE in 1h 55m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8267c193cbe438cef45cea29c873b5703d6a5013","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-30 15:18:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/2/check/DellEMC_Unity_FC/815ea95/DellEMC_Unity_FC/86 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/2/check/DellEMC_Unity_iSCSI/919ead4/DellEMC_Unity_iSCSI/76 : FAILURE in 18m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cafeecde63d67591e19a59b860b7f17bef17df57","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-30 16:34:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/63/788363/2/check/DellEMC_SC_FC/b198399/DellEMC_SC_FC/51 : SUCCESS in 2h 00m 14s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/63/788363/2/check/DellEMC_SC_iSCSI/960811c/DellEMC_SC_iSCSI/48 : SUCCESS in 2h 04m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"cfbc2568361f019b0baa73bb9300a682825c1bad","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-30 16:37:22.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-2/Lenovo-iSCSI : SUCCESS in 50:22.62\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-2/Lenovo-FC : SUCCESS in 46:31.17","accounts_in_message":[],"_revision_number":2},{"id":"901710c66ebed666f87460c8dc9b515e28058d88","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-30 16:46:06.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-2/HPMSA-iSCSI : FAILURE in 1:12:49\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-2/HPMSA-FC : SUCCESS in 34:00.23","accounts_in_message":[],"_revision_number":2},{"id":"e4e3b431ec603de082600be9cf6a63b6f3f525b4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 16:55:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/63/788363/2/check/iscsi-driver-nimble/30d16c8/30d16c8.html : SUCCESS in 1h 51m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6fd458112e7c8216e50940a077bbcc2e3de313bc","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-30 16:59:01.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-2/PVME-iSCSI : SUCCESS in 1:03:36\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-2/PVME-FC : SUCCESS in 56:01.74","accounts_in_message":[],"_revision_number":2},{"id":"013a390e30c45cb97859ece45fa2d1c47fd71528","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 17:03:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/2/check/DellEMC_XtremIO_FC/9456c82/DellEMC_XtremIO_FC/95 : SUCCESS in 2h 03m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/2/check/DellEMC_XtremIO_iSCSI/45a658e/DellEMC_XtremIO_iSCSI/95 : SUCCESS in 1h 51m 48s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"90dc96cb90fbc762bf8fadb64796db2b44697e29","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-30 17:20:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/2/check/cinder-storpool-tempest/c0bfd50/ : POST_FAILURE in 2h 17m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5a2ec5a7cbdb9a162562abc5a3320487dc6aac18","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 17:33:21.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/63/788363/2/check/DellEMC_VNX_iSCSI/d03546a/DellEMC_VNX_iSCSI/55 : SUCCESS in 3h 02m 59s\n- DellEMC_VNX_FC http://publiclogs.emc.com/63/788363/2/check/DellEMC_VNX_FC/806175c/DellEMC_VNX_FC/58 : SUCCESS in 1h 15m 17s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c06e6db3fa77736095275145da05cad422dd42c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 18:30:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/63/788363/2/check/DellEMC_PowerFlex/a2021e3/DellEMC_PowerFlex/30 : ABORTED in 4h 00m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"22f67acca8e23b525c1bc24538045dc61e4e751e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-30 18:55:57.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-2/ns5_nfs : SUCCESS in 4h 58m 7s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-2/ns5_iscsi : SUCCESS in 4h 48m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4f8e948deebdf50dfcf5691fee7e3df2cd170744","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 23:42:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/63/788363/2/check/DellEMC_PowerMAX_FC/8880eeb/DellEMC_PowerMAX_FC/40 : SUCCESS in 5h 44m 55s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/63/788363/2/check/DellEMC_PowerMAX_iSCSI/81c4353/DellEMC_PowerMAX_iSCSI/34 : SUCCESS in 5h 25m 05s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"47f0a6fe3cbb15f93e61aa837d233c0fcb073895","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-01 00:16:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/2/20210501065955/NEC-ISCSI : SUCCESS in 2h 13m 48s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/2/20210501065955/NEC-FC : SUCCESS in 2h 16m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e6ad082827f53e4dd2cfe08dfbd9185fa797f69f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-10 18:25:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0b11a1cdaab2576a05fd749156f7d2c605721fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-10 18:28:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a83e8ac390caf517ddfcf30295574b2e48ab2751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 18:54:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eaae9a4ee42433f9a3217dbfc1eb15e : SUCCESS in 22m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2542b5e075a248a19f4a5740d6f50e89 : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7fdf81212d8cc4623865a7c1cea107887b93a6a3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-10 19:24:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/3/iscsi/ : SUCCESS in 54m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/3/smb/ : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":3},{"id":"2f7869638a1af61f0a4318906b19ac571eaaa08d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 20:14:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c8cd1f075c04d4c9b19433ff66e984e : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c04d96e7a1a4453284d1c68078f10f24 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f713fc0fe8f64b2a871429d4026dc9fb : SUCCESS in 12m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f626a25d5524e66b12b268b591ead7b : SUCCESS in 9m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbe37e482b554028803d96b23020e4e9 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fdde0939d344f5e91e45739f0ea9703 : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/474a913ef72f458daed07cf9d0c290c5 : SUCCESS in 57m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0de5209e33bc48bc9652b00f2edae2f5 : SUCCESS in 58m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e55b9fc7c1de4f27b188ecdca2d0ad2f : SUCCESS in 15m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b46d2e2baf3e4f68b82d5363f7ff7503 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7043f192144b4deabebbea9802ee1af4 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0846419e7e0945ab8f8ac1043f79ec26 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd4d20ce6167434eb5f55676490e7856 : SUCCESS in 35m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e16308b4c73749a0860764fc6bb2d970 : SUCCESS in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7f1aed91ff9b4d16b68d1be219a76657 : SUCCESS in 1h 18m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2433973ebabd4dd19790116ff0d45cf3 : SUCCESS in 1h 37m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0b36ae6abfe4242b16848dd25bdea21 : FAILURE in 1h 11m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9a62f13619f246d39c02bb2cea20661d : SUCCESS in 1h 28m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/98c045ef2cb643a8a4cc9783f6d64934 : SUCCESS in 58m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/46badebf32e94966a2158b141d59245f : SUCCESS in 1h 30m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ba4b87c07b74251bcbf22755bb331f4 : SUCCESS in 1h 44m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d46172ebeb2481a9138420d7154f7b8 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":3},{"id":"02e82f2d522ee6287566463ec35672b9efad8893","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-10 20:43:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/3/check/cinder-storpool-tempest/a82f05d/ : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f036d5b41c6b0cfae19c56acc1024699d33aabac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-10 21:04:00.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/3/check/tempest-dsvm-full-focal-py3/8ccbaed/ : SUCCESS in 1h 45m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/3/check/devstack-plugin-ceph-dsvm-focal-py3/6813628/ : SUCCESS in 1h 38m 38s","accounts_in_message":[],"_revision_number":3},{"id":"6aab3128e263d1dce3c6547422d8f5f585ff7b00","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-10 21:30:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/3/check/dsvm-tempest-cisco-zonemanager-job/14d3485/ : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3d5d95a4e366b6a629cb3fd6109822b26a6076a5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-10 21:37:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/3/check/iscsi-driver-primera/0d0b94c/0d0b94c.html : FAILURE in 58m 17s","accounts_in_message":[],"_revision_number":3},{"id":"2594362ebf860d98b7a12674d2ecf93070abc1fb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-11 00:26:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4e0a872/ : TIMED_OUT in 3h 02m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ae27702/ : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6b850aa76e54cfce669d9ff169f0f9dff2442cb3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 02:57:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c4b8a91/c4b8a91.html : FAILURE in 52m 40s","accounts_in_message":[],"_revision_number":3},{"id":"6ec373566aa6916ec7b6145e4ee0bfdc0cc66298","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-11 04:51:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-3/ns5_nfs : SUCCESS in 5h 55m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-3/ns5_iscsi : SUCCESS in 5h 36m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"1e9d7184c5a4353c74e58dbcc533698dc7b5311a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 06:25:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/63/788363/3/check/iscsi-driver-nimble/482c698/482c698.html : FAILURE in 25m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5f9fc475fdc5028627ea8f46667ddc24140452b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-11 08:07:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/3/check/DellEMC_Unity_FC/e27cb49/DellEMC_Unity_FC/149 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/3/check/DellEMC_Unity_iSCSI/85017c6/DellEMC_Unity_iSCSI/137 : SUCCESS in 1h 31m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e97eefcb70c7cc1ff358150b3db1059e7035e073","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-11 13:57:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d3956e4a4812903fecb4fc0eb1995b60f1d30e03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 14:19:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7332d3c86d1a487ab8e90615464231b6 : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4110371adc9c4925bf6071e3f9e0d018 : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c45df7f9199549777a30990439137860d965bda1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-11 15:07:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/3/iscsi/ : SUCCESS in 1h 09m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/3/smb/ : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":3},{"id":"76fe08c070626c2f3133bd68f814aefd8d5cf866","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-11 15:15:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/3/check/dsvm-tempest-cisco-zonemanager-job/ade6ba1/ : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":3},{"id":"3941779341403b8f9b2371bc7f2e67ea2884c0fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-11 15:34:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/67b3681/ : SUCCESS in 1h 36m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d8f78f8/ : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":3},{"id":"402488da0a740cdde7b8c720bc1346b05e05ef84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:59:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5d54342f9814821a72251fd5560a661 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a666d0966d45138562befa5d7f8827 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/922ad4c137ae4a47adda01c5dcfdaba6 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55890fdc8b5d43bd9dca1f8364d1e80f : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04fe3f033bee43cbb06c460659f1f3ef : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3108b800ae4f61985e1d8f680e17e1 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6d2b6811434c4315868e4ca378a783a2 : SUCCESS in 52m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c84f54cf0dc045579aec1b690e8b3be7 : SUCCESS in 1h 10m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8339d7e3c4a042148a595b2ad9d9722d : SUCCESS in 11m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d661ca23a0f54f5fa9637bc8f35c2f1a : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a74a88b6afd2494a8da1992ebe23d132 : SUCCESS in 7m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5af3c06cbb29499ab72f15c9ee4452f8 : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bfc1675d56d3496aa7c6b9473bb3d892 : SUCCESS in 45m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec04e14e35774dbfab1140374dda11d2 : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/460f84a97a224d999e35180c9e799930 : SUCCESS in 1h 25m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/51db5aaff816456bb6ee2ce29a9afaee : SUCCESS in 1h 38m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1b5ea1c375649e2b74297c6fd51ff7f : FAILURE in 1h 41m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/587fe256960d40dfbbfb7132dfa638b7 : SUCCESS in 1h 14m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/79e61a26442a479a9ffff74cb7aaaf14 : SUCCESS in 1h 13m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0348a6276cf418e870ccb0f821163fe : SUCCESS in 1h 20m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b95d78d7c8f4400bc98941e7f3aa5dc : SUCCESS in 2h 00m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70ead5a2e1e74766ae8fbf13f60a6cff : SUCCESS in 47m 19s","accounts_in_message":[],"_revision_number":3},{"id":"efcdc9d5e9cc2b9e4fb36b7ee2cf9b3a4ecb42d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-11 16:19:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/3/check/cinder-storpool-tempest/3123ff6/ : POST_FAILURE in 2h 19m 39s","accounts_in_message":[],"_revision_number":3},{"id":"53ed141d6264dbe844e8ec1b2db1bc544640b1bd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-11 17:28:08.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/3/check/tempest-dsvm-full-focal-py3/90e79fb/ : SUCCESS in 1h 37m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/3/check/devstack-plugin-ceph-dsvm-focal-py3/c1f8505/ : SUCCESS in 1h 43m 03s","accounts_in_message":[],"_revision_number":3},{"id":"55d9fbec01f91b45ed6d75508b01017e19ebffae","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-11 21:25:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/3/20210512034035/NEC-ISCSI : SUCCESS in 2h 33m 47s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/3/20210512034035/NEC-FC : SUCCESS in 2h 44m 56s","accounts_in_message":[],"_revision_number":3},{"id":"2c669a12b194e3abd98eac25620caa00cb1279fe","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-17 13:16:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1374386f8e1f1cc223c65b6a507ed3875570727f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-17 13:48:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/3/check/cinder-storpool-tempest/24d42da/ : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":3},{"id":"93b7f9312de49bbcf149346dd693784f94804ce5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:59:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5101d311b9b456583e4df559927baad : FAILURE in 33m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49a1c1b0328b462787a77602708c1809 : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29c6184cfd4541d2ceef6fc3ac211f797654788a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-17 14:10:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/3/iscsi/ : SUCCESS in 53m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/3/smb/ : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":3},{"id":"662a1660c0d3fa707daa1d355542aeabebf2e222","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-17 14:52:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/3/check/dsvm-tempest-cisco-zonemanager-job/e1ab622/ : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f86c9c44e00ecaed4659adfd2bc4df2b5439d74a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-17 15:01:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/da1ff6c/ : POST_FAILURE in 1h 37m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4af9dbd/ : SUCCESS in 1h 43m 52s","accounts_in_message":[],"_revision_number":3},{"id":"061d3732c79528d7759fc9d9fc863890b76197c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 15:26:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b28ceb55080f43a68ec4d94d6950fe94 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997eeed07cea4f728209ef038f3eae36 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ca50dc252074483b70b570901990dfa : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f5e11d6c04a487ea4c0947dd6c928eb : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56b59ce9e6649c9821176e6a119ac5a : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befc367cb4b74798b7ab7e7f8ffb786f : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7ef40ff60a464f51bf3100a8c75f3b63 : SUCCESS in 57m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05300570271e49609abb46bc09ea0956 : SUCCESS in 1h 00m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32a6e5e68a43428b968f584160bc7ec2 : SUCCESS in 12m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4f05aa224994b059688a5553f4d6137 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/044737306ed445e1b1fdb1a6fb139023 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6a63f8093c240dcbe2786f8b6327c72 : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed8672489c3a440bb3f80884ebe7b685 : SUCCESS in 35m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/659c20a0a82942209e92fe4f78c90c66 : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/21b19ae305f74457a86cd15be07c0193 : SUCCESS in 1h 08m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3c78dc6ce0ab45e6b108f588de96943c : SUCCESS in 1h 16m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8497ac88228c4c4893ca089b0e4ef1ed : SUCCESS in 1h 18m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5bd516417f4b48c2baef7010a2954e59 : SUCCESS in 1h 16m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6e5f281dc53a42fbb0053077baa7cfd6 : SUCCESS in 1h 13m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a1b72045a3d842339559eeab9673e6a3 : SUCCESS in 59m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a57db23934845b6b4e97f54a860559a : SUCCESS in 2h 03m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d83536b8ae17424ab5af58ad981d96a3 : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":3},{"id":"060e19d741c86b0bb47024c0c293faeb045c4318","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-17 15:52:28.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/3/check/tempest-dsvm-full-focal-py3/3930974/ : SUCCESS in 1h 38m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/3/check/devstack-plugin-ceph-dsvm-focal-py3/a0083cc/ : SUCCESS in 1h 44m 13s","accounts_in_message":[],"_revision_number":3},{"id":"50c287aabeca67b2948bca87ec6baf090c76d42e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-17 16:30:59.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/3/check/DellEMC_PowerStore_iSCSI/8f73011/DellEMC_PowerStore_iSCSI/94 : FAILURE in 1h 01m 35s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/3/check/DellEMC_PowerStore_FC/a4e61e4/DellEMC_PowerStore_FC/95 : SUCCESS in 1h 03m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"02d1cfc2ddc15c70ca66ce917ed1ce3f7d25386e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-17 17:19:21.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-3/Seagate-iSCSI : SUCCESS in 46:07.69\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-3/Seagate-FC : SUCCESS in 38:15.98","accounts_in_message":[],"_revision_number":3},{"id":"c618c76f6025af1ca9829b2bdd94c4ca323ce9dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-17 17:38:08.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/63/788363/3/check/DellEMC_PowerFlex/ea6ec98/DellEMC_PowerFlex/103 : SUCCESS in 1h 43m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"683bddc3ab261c03c59116acbbe94e3fa82a6530","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-17 17:38:38.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-3/Lenovo-iSCSI : SUCCESS in 48:02.33\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-3/Lenovo-FC : FAILURE in 54:49.78","accounts_in_message":[],"_revision_number":3},{"id":"5da97fd68f1304da2a6993b553e19143c01c3b68","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-17 17:45:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/3/upstream-check/cinder-cDOT-iSCSI/55c1baf/ : SUCCESS in 2h 09m 51s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/3/upstream-check/cinder-cDOT-NFS/d9806fb/ : SUCCESS in 2h 18m 48s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/3/upstream-check/cinder-cDOT-FCP/63b2ff1/ : FAILURE in 1h 05m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"bd1ae4f0819ca2106fad501b97a8ee2f2d75caa9","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-17 18:14:37.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-3/HPMSA-iSCSI : FAILURE in 1:26:29\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-3/HPMSA-FC : SUCCESS in 33:24.28","accounts_in_message":[],"_revision_number":3},{"id":"ba50856408771cc05274274160d3d40954c7c75c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-17 19:02:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/63/788363/3/check/DellEMC_SC_FC/6a7cf3e/DellEMC_SC_FC/110 : FAILURE in 2h 10m 27s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/63/788363/3/check/DellEMC_SC_iSCSI/bb67854/DellEMC_SC_iSCSI/107 : FAILURE in 2h 52m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"07875bb875044b02a0a27476b7dbd42e047e362a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-17 20:11:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/3/upstream-solidfire-check/cinder-solidfire-iSCSI/2b33116/ : FAILURE in 56m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"54c0645f80342303c767a60e307cc8adb3bc05b9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-17 20:51:43.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/63/788363/3/check/DellEMC_VNX_iSCSI/ee48edf/DellEMC_VNX_iSCSI/115 : SUCCESS in 3h 40m 49s\n- DellEMC_VNX_FC http://publiclogs.emc.com/63/788363/3/check/DellEMC_VNX_FC/da77af3/DellEMC_VNX_FC/121 : SUCCESS in 1h 40m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"560d29cccdf5c4fa280b83921090da8043b76e12","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-05-21 15:54:51.000000000","message":"Patch Set 3: Code-Review+2\n\nThis looks ok to me.","accounts_in_message":[],"_revision_number":3},{"id":"8054145ea0a6b3008a069ca12ba094911b8974a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-23 15:25:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ac908d3278eb7ba9ef2097ede2623be0c8dd8e45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 15:48:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3054ec4cdfa413a9dceea9329784af4 : FAILURE in 19m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de0c6253bc27485d995b20c3f64bda12 : SUCCESS in 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"85673bae62b06c481c4cebc3621769b3206925b2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-23 17:02:21.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/4/iscsi/ : FAILURE in 1h 35m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/4/smb/ : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":4},{"id":"26a14fa559efbd5b8294f64d8425f15ec97b24df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-23 17:03:05.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"2dc3a2c01c01c23bd670f47dd655000d1d10fd39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 17:41:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecf16067731e451c89d7319c6433161f : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5a70400a1e47c1b2a8f19fb7915680 : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bdd79beee344dafa92fa9f5967e8f58 : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e21a0aeba614aecbfc882df9c3e1e17 : SUCCESS in 11m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31c4c8e3c5e340ef9536a2ac42323242 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe4022a179245b382cecd437dc28722 : SUCCESS in 15m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/2728b2af4cde4739977e31340abe0cbc : SUCCESS in 1h 12m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dcbef129e2054627b03f68b043e1c95c : SUCCESS in 1h 13m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08fcbc0cbe76486c883467d7a080863b : SUCCESS in 16m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/689ca2ff3add45068d0c9b325df83b72 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6754909007e1483c9ec03239671016f0 : SUCCESS in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b97d55a003af42b6a57caaa46f199cab : SUCCESS in 9m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0825e100eeaa40da9f1a5e56076cbfd5 : FAILURE in 33m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9ee9e25af384c86bc168c058b28ac9b : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7398b17e026c4d809c681b35344e47fb : SUCCESS in 1h 20m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d5a2512892e4d0ab6bb4fc93768fe79 : SUCCESS in 1h 26m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef0d8f4c921d49a09fea6900b193a4ba : SUCCESS in 1h 27m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/505816def4024e8dac8a44e2d8baeb19 : SUCCESS in 1h 10m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9444537c46964d1d97369c05eae06c18 : SUCCESS in 1h 15m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/50c85d9c59d34877aa38b2f05a412921 : SUCCESS in 1h 23m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/baa3f24531354f64bfd78074ea17736f : SUCCESS in 2h 07m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec996a9168964e63a83c1438bba52a40 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":4},{"id":"b233de591d153a1e5e24d7dccbeb95b11281b748","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-23 18:22:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/63486e5/ : SUCCESS in 1h 43m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/eef2054/ : SUCCESS in 1h 37m 28s","accounts_in_message":[],"_revision_number":4},{"id":"6646bf17d638965f18f4f0519748ebdad17969f7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-23 18:49:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/4/upstream-solidfire-check/cinder-solidfire-iSCSI/ebaa346/ : SUCCESS in 1h 06m 40s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"3dd3abf7b330a30ad348708bfc40fd7cf57e5349","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-23 18:49:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d805d39/d805d39.html : FAILURE in 53m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b2f459ffd7f099a94617ebb941427baddfa484b7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-23 18:59:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/4/upstream-check/cinder-cDOT-iSCSI/1fdfaa6/ : SUCCESS in 38m 10s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/4/upstream-check/cinder-cDOT-NFS/2b439b9/ : SUCCESS in 36m 38s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/4/upstream-check/cinder-cDOT-FCP/10aa63b/ : SUCCESS in 1h 16m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"e422cbdf9558d0f3a0f5d154ce92610d418f7819","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-23 19:22:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/4/check/dsvm-tempest-cisco-zonemanager-job/b976f90/ : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":4},{"id":"d262719a92dae89bac7e51126374187dbba55eb7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-23 22:28:41.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/4/check/tempest-dsvm-full-focal-py3/d97bc68/ : SUCCESS in 2h 00m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/4/check/devstack-plugin-ceph-dsvm-focal-py3/0e3dd5b/ : FAILURE in 42m 01s","accounts_in_message":[],"_revision_number":4},{"id":"58139143f968b567577cf9825355932445d851f1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-23 22:31:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/63/788363/4/check/iscsi-driver-nimble/717b827/717b827.html : FAILURE in 1h 45m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f0dfb5231969d1d8e6da8b3882a0e8974b4475d1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-23 22:57:16.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-4/ns5_nfs : SUCCESS in 4h 36m 28s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-4/ns5_iscsi : SUCCESS in 4h 41m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"235fad860d11af41109b063cf59beb84c9de457b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-23 23:24:45.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"274cd539810b19de57059af013c53fcad5b99587","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-06-24 00:35:15.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-4/Seagate-iSCSI : SUCCESS in 41:11.21\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-4/Seagate-FC : SUCCESS in 36:48.37","accounts_in_message":[],"_revision_number":4},{"id":"554a6515edadacbee79a7101eb863f08fce4c1e5","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-06-24 00:47:25.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"65780040765438067f703765da1199a884477f65","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-24 02:43:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/4/check/DellEMC_Unity_FC/f8a66db/DellEMC_Unity_FC/338 : SUCCESS in 1h 20m 17s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/4/check/DellEMC_Unity_iSCSI/c101bfa/DellEMC_Unity_iSCSI/320 : SUCCESS in 1h 40m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"060f3f574f77bde4413a421b49de4ca1b4f04a6d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 02:54:34.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/63/788363/4/check/DellEMC_PowerFlex/bc89323/DellEMC_PowerFlex/213 : SUCCESS in 1h 48m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f77004be61d39a805e5df069daca2b688d6914b4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 06:52:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/4/check/DellEMC_XtremIO_FC/bfdc0f1/DellEMC_XtremIO_FC/388 : FAILURE in 1h 20m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/4/check/DellEMC_XtremIO_iSCSI/3512a8c/DellEMC_XtremIO_iSCSI/385 : SUCCESS in 2h 18m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"229f361545849689adc37933299403a84c9328fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-24 07:39:26.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788363/788363/4/check-not-vote/ext-cinder-zuul/43c2e4f : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":4},{"id":"6bceb76b843fa780f20037978dd6d8bbb23757f7","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-24 09:17:44.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-4/Lenovo-iSCSI : SUCCESS in 48:34.22\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-4/Lenovo-FC : SUCCESS in 45:12.60","accounts_in_message":[],"_revision_number":4},{"id":"a36a74b00a29c52df00a0220b64488e2f2663bd0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-24 09:18:42.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-4/HPMSA-iSCSI : FAILURE in 1:00:03\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-4/HPMSA-FC : SUCCESS in 31:29.53","accounts_in_message":[],"_revision_number":4},{"id":"ed296eaf7a92459f5ad5f7cab5ad877b831a7809","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-24 09:26:07.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"15c9adfff502606c236c04380c4442246d3537da","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 16:45:28.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/63/788363/4/check/DellEMC_VNX_iSCSI/444076d/DellEMC_VNX_iSCSI/186 : SUCCESS in 3h 53m 02s\n- DellEMC_VNX_FC http://publiclogs.emc.com/63/788363/4/check/DellEMC_VNX_FC/9b60b47/DellEMC_VNX_FC/193 : SUCCESS in 1h 37m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"914992f1444ac34fa009e30ed0ffb3f3124d4dc2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-24 18:05:30.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"12ff90c3c5cf4d70d401db0a2f3a8113c0444109","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-25 06:53:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/4/20210625130714/NEC-FC : SUCCESS in 2h 34m 45s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/4/20210625130714/NEC-ISCSI : SUCCESS in 2h 45m 52s","accounts_in_message":[],"_revision_number":4},{"id":"3edf7b9a945ff7d558b0b5420aadbe7f3190feed","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-30 20:44:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd850e09a7f8aa0a2d3963e14d9726a5cf280a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-30 20:45:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"da0226bf9e4ec7c0a256a544e8a5d9d0f0eb3794","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:06:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6140c494319643f88ebaf70a2bab330f : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98b5ed4cd5464b43a4e8beee9b26ba2d : SUCCESS in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4ed0af3455a3ccf3e9aecb4e294b1ccb4d3f2637","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-30 21:50:18.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/5/iscsi/ : FAILURE in 1h 03m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/5/smb/ : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":5},{"id":"daa36fd0d77c82cb00672f156dd2a089ec1d18e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 22:49:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b865cedc5bc46859cff87c95343ca39 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797228400a0949a2968b9d4aab77195a : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384eff71c4e9498caaddc2e9bf6eb2d3 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91b4fa876bed48c68cbfc8338ffb62a5 : FAILURE in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb85e14d4bd44e39a608e704eb97de73 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0632428d774402ba80d16a3b8c34a6 : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f134e290aeaf46bc88904cb488734f35 : SUCCESS in 55m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f349dd604f2c4909babe9a31914c468d : SUCCESS in 1h 02m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/890f5d727a9e40e8b646bebccddb6007 : SUCCESS in 12m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c5882ec957b4fdfb1de33add8595396 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/719f4ebbc7f54b269c0bbc96c8305460 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0312aaa235794132b50e8190d675c981 : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5017fb9ecaeb4037a11544403bb73a19 : FAILURE in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a0583024637490096c55e59be39b775 : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b889e1d337147f7b5232c88683bd3df : SUCCESS in 1h 21m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/94b9ada384c2450ca97ce4bf775f4094 : SUCCESS in 1h 23m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c05daaaa34ac45d6a12ef713816db335 : SUCCESS in 1h 23m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3e2745000a542e5bd6272dedb43d349 : SUCCESS in 1h 18m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1699674295c40bb8d974202c569bb99 : SUCCESS in 1h 13m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aeb655f70ce442858e9294fa3d8296bd : SUCCESS in 1h 18m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db226e54e4c7495a9205f151f31a3064 : SUCCESS in 2h 01m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c620afb7e384a4c87d880a213092c49 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":5},{"id":"8a7ff71b437c3a9ccc6bd801ee62ad485806e774","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-30 22:49:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e239815/ : SUCCESS in 1h 33m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/210e115/ : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":5},{"id":"da16c09ca30c8ac1bac5d85acd2a292d510695eb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-30 22:53:42.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c83671fa38ac4e3cb54bde750f8bcde8 : POST_FAILURE in 1h 45m 36s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2d6cfd004a9e4c829f39df5575d711ce : POST_FAILURE in 1h 10m 54s","accounts_in_message":[],"_revision_number":5},{"id":"636b13c630583779ae5010ff27a908f90f0fd7c1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-30 23:01:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/5/check/iscsi-driver-primera/2ec3a73/2ec3a73.html : FAILURE in 53m 01s","accounts_in_message":[],"_revision_number":5},{"id":"6d36defed3ad285092a49fdccfac2b8e621a1d17","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-30 23:02:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/5/check/dsvm-tempest-cisco-zonemanager-job/f906cd2/ : FAILURE in 1h 05m 46s","accounts_in_message":[],"_revision_number":5},{"id":"988e7cc7757cd318a13d638034b41e9b625abdee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-30 23:52:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1bacd62/1bacd62.html : FAILURE in 51m 33s","accounts_in_message":[],"_revision_number":5},{"id":"c7ea8dfad911cb5078c306126d95db2393079c2c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-01 03:16:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/5/check/DellEMC_Unity_FC/601c204/DellEMC_Unity_FC/387 : FAILURE in 1h 50m 28s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/5/check/DellEMC_Unity_iSCSI/d809352/DellEMC_Unity_iSCSI/363 : FAILURE in 1h 13m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e6941795df4cd87dfd5ceef75aaaa56240a0ec3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 03:18:22.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"f94307a6c59cd393562954e9b1b4c2f1986d4d34","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-01 03:47:34.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788363/788363/5/check-not-vote/ext-cinder-zuul/eb32f81 : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":5},{"id":"12202f82f7a117ed175e42e3348228ef605c8339","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-01 09:31:36.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/5/check/DellEMC_XtremIO_FC/f11ab35/DellEMC_XtremIO_FC/430 : FAILURE in 1h 18m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/5/check/DellEMC_XtremIO_iSCSI/bc40bd2/DellEMC_XtremIO_iSCSI/423 : FAILURE in 2h 53m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"e7ce0cda40055b5565c9e48f4cd2ee5a8be3433c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 10:50:05.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"651db78f696dd61b9e70d002890b8ea0e5fd3631","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-01 18:39:29.000000000","message":"Patch Set 5:\n\nrecheck\n\nrbd unit test failure","accounts_in_message":[],"_revision_number":5},{"id":"4c786c2aef706e87e1328366d8f767ef24b42568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:17:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d46e2ecddcdd473b9ed2e0be3c3a66d8 : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ce63aa11bd049eaa05a91926315b83c : SUCCESS in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d9ed8f996d98ff7fe92234f84054bd38c844bf33","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-01 19:35:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/5/iscsi/ : SUCCESS in 54m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/5/smb/ : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":5},{"id":"e47cd1b6e300328c1be7543ab9cdf215901b2fd8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-01 20:09:18.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/3384d40d25ec48bca631fc2635fb5189 : POST_FAILURE in 1h 23m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/3fa599b601094646b9aba938c7d8c0b7 : POST_FAILURE in 38m 34s","accounts_in_message":[],"_revision_number":5},{"id":"2bb32f6e7abdb536a4316a227e908e85312e1b2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 20:10:15.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"b2497ef3a5bfffdc5f15bfd8a445291a1780e7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 20:36:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3393b53af9e4151b2e3296a87900bd6 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf3a692378743008821abe33c42d6af : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7c239e9fbd54cdaae6b820e0bffb88e : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f85ed07ce37453a9e83c471f4ac0884 : SUCCESS in 12m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a96b411ea9f849b4b67fd3c739dcdead : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d035b6dc5dbd4b61a279daf6a3e6fbf8 : SUCCESS in 22m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/e9baa310e6034573a47feb03366a1f25 : SUCCESS in 57m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b73bede8c8e4d0c8714d7ab6adfad51 : SUCCESS in 1h 08m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/87940beda65d41f6a64330e095f677ae : SUCCESS in 13m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/624aad9f13af45559eb8eea27c2aac6d : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/147afae2d8a5471cbd3888b0c724b445 : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a13e88f6833141a5898ae9affdc275d4 : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd0a4dea24e5460ba74deb0cf5b0b472 : FAILURE in 34m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a40f64c9f2943b18353b37bc46a0504 : SUCCESS in 7m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5aa23b58904a435bbd43f32b562e22d8 : SUCCESS in 1h 18m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2fd1f285badf445c9cbc4fc18eacc5ca : SUCCESS in 1h 47m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f33f261b7274b2b896a09249a3c0a61 : SUCCESS in 1h 23m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e28ac7ee0c84ea9bbd5a605b6f6abd9 : SUCCESS in 1h 08m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0866a6b61e184551b77cd179a24271d4 : SUCCESS in 1h 15m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a6aec3508bbc40308d5e5c67becc9623 : SUCCESS in 1h 06m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bbed40a5b5e4482b515f8e0b67f313a : FAILURE in 1h 53m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70026cdec7f040bb8b5366ea0f15457f : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":5},{"id":"6bc1d1f338aa8b9b96784036a94f3b6e4bfef48b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-01 20:59:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/5/check/dsvm-tempest-cisco-zonemanager-job/e34b5f3/ : FAILURE in 1h 04m 08s","accounts_in_message":[],"_revision_number":5},{"id":"08e3c9858f70f47f26fb147a68135029540b58a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 00:13:26.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"1c366be3038ab59cab4490c69526e98442533def","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-02 02:14:16.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/13895/ : ABORTED in 50s","accounts_in_message":[],"_revision_number":5},{"id":"06591d317a5fad1b0b1907d387bf32fdde9812b1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-02 09:54:44.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-5/ns5_nfs : SUCCESS in 4h 59m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-5/ns5_iscsi : SUCCESS in 5h 1m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"de8993187900a1f06cbed908e89c87e8f5003f2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 11:05:21.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"6d36760e490a0a551f37800380376e745a3f4f21","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-02 16:26:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/5/20210702203847/NEC-FC : SUCCESS in 2h 54m 17s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/5/20210702203847/NEC-ISCSI : SUCCESS in 4h 48m 03s","accounts_in_message":[],"_revision_number":5},{"id":"29a5bfd167134cd46ee671ca33476b2c5ca5b0ee","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-06 14:15:55.000000000","message":"Patch Set 5:\n\nrecheck\n\ntempest.scenario.test_network_basic_ops.TestNetworkBasicOps.test_subnet_details","accounts_in_message":[],"_revision_number":5},{"id":"f5ddf5e10d9548674c44812c3128db5598efb36c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 14:40:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da31885c4b9741738840883355c59401 : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a88a9821e7124143b3bcbef4acf9d953 : SUCCESS in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"49c10f107741a90c0903bb9dc182661edcdbce53","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-06 15:10:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/5/iscsi/ : SUCCESS in 54m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/5/smb/ : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":5},{"id":"7217cf6574a54f6726ca5d13b53a4fa8045c92f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 15:26:15.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"2b375afe0a2a4967cfd5751e22b748719aa60dbe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-06 16:29:39.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/5/check/dsvm-tempest-cisco-zonemanager-job/96cf136/ : FAILURE in 1h 04m 46s","accounts_in_message":[],"_revision_number":5},{"id":"6949c56e42a3b16dfb39048f074dd7d534e89cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 16:47:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94e859a25e6a4dc39ed5a7b271ea580e : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651732f7be1d4a26a7e34423faf098d2 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e1573d4e747408b9a067a2cc9fdc3d9 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25d59701c07841468e41466a1276a7ed : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2deee566013d4b1abe78b47c893cb7d7 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf032095a83f42b5b4d55cd5473ab6b6 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/1c00f2e1d1f44b1c838ce784e4aa781c : SUCCESS in 1h 03m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c58fddae3ee14a07b28fb61e9057d0fd : SUCCESS in 1h 15m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a6d43c66964143c3a95dcbf6001b22bb : SUCCESS in 12m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4ec5876dbc247529887fa04c40cd05b : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4ec8eaa9cdb4db591a6e249986066f4 : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3baa20bdb7db43c9a90e2200fd38925c : SUCCESS in 7m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dea6bb80292548e4b17d87c1641b1ed6 : FAILURE in 35m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acc26542e9fc469ab1a3876df762f336 : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f74c940b3d144ad8d2f24af2b90efa7 : SUCCESS in 1h 21m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b040f3bdc20c400bbb0ebe63d6a012fc : SUCCESS in 1h 35m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf7db7ce2815435592e781fdd7b35609 : SUCCESS in 1h 35m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89cbfe461ff547e4b9898ea3ac1d91d0 : SUCCESS in 1h 13m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6ec195b2a60c4f848e9ffbd2a00f7b2b : SUCCESS in 1h 19m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e3303490bde74986a3a857399d2e9524 : SUCCESS in 1h 26m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad5fb318838744258987190b4c3346c9 : SUCCESS in 2h 25m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79eeae4f5800475ea899c2fedb717933 : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":5},{"id":"4a7026f5a67c31344113cd2d840f30da12432ba5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-06 17:46:33.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/5/check/DellEMC_PowerStore_iSCSI/16b6bc0/DellEMC_PowerStore_iSCSI/205 : SUCCESS in 57m 52s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/5/check/DellEMC_PowerStore_FC/139b66d/DellEMC_PowerStore_FC/206 : SUCCESS in 52m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"d96952d58263bbaa73ca51bdf392b6a646497db2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-06 17:46:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/5/upstream-check/cinder-cDOT-iSCSI/71cb53e/ : SUCCESS in 35m 16s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/5/upstream-check/cinder-cDOT-NFS/82120b2/ : SUCCESS in 36m 00s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/5/upstream-check/cinder-cDOT-FCP/728c062/ : SUCCESS in 57m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"02b799097ddde7c8d39e3b6387481ff6f4034fb4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-06 17:57:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/63/788363/5/check/DellEMC_VNX_iSCSI/8df8262/DellEMC_VNX_iSCSI/235 : FAILURE in 1h 08m 38s\n- DellEMC_VNX_FC http://publiclogs.emc.com/63/788363/5/check/DellEMC_VNX_FC/771ea55/DellEMC_VNX_FC/253 : FAILURE in 55m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"3b8ba858eb40e500e010ca93e310b5361d23a99a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 17:57:21.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"6508dde7dd595a2934d05cea8729388fb9146ab6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-06 18:10:11.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/19896e89a96e4f33afc1b136ab7722fe : DISK_FULL in 2h 48m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"a4fdc1311b5470144e0d204fa73684787347c17a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 18:10:44.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"754431f7dbb6f9abe80808906d4bd5ce5ed4052b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-06 18:23:15.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/63/788363/5/check/DellEMC_PowerFlex/6b72f8a/DellEMC_PowerFlex/274 : SUCCESS in 1h 34m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"77c8e197bd34b39439eb6dc14566684013b872ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 18:23:41.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"f52e6fef5cbc7e7279ee2556029b7a79b0ed933f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-06 18:33:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/5/upstream-solidfire-check/cinder-solidfire-iSCSI/c0d0ab4/ : FAILURE in 1h 43m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"d6f45a4a9e1d629097d4bfa34251ca52b0546db8","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-06 18:37:44.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-5/Seagate-iSCSI : SUCCESS in 41:04.95\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-5/Seagate-FC : SUCCESS in 37:00.60","accounts_in_message":[],"_revision_number":5},{"id":"bd08967ed09d7ec620ddf29a9686f8db9a898325","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-06 18:50:04.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-5/HPMSA-iSCSI : FAILURE in 1:00:18\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-5/HPMSA-FC : SUCCESS in 30:43.02","accounts_in_message":[],"_revision_number":5},{"id":"33b2df24c8c10f5fd2c58c0f2c7740ce929c602e","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-06 18:57:55.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-5/Lenovo-iSCSI : SUCCESS in 48:40.37\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-5/Lenovo-FC : SUCCESS in 48:07.86","accounts_in_message":[],"_revision_number":5},{"id":"069742c756528e6633ba26234ca7c29dd44f16f2","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-06 19:14:23.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-5/PVME-iSCSI : SUCCESS in 59:14.83\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-5/PVME-FC : SUCCESS in 53:02.75","accounts_in_message":[],"_revision_number":5},{"id":"1029fae407a1bb62f91f68c204d8caf2d48dfad1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-06 19:29:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/5/check/DellEMC_XtremIO_FC/5fb775e/DellEMC_XtremIO_FC/446 : SUCCESS in 1h 18m 50s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/5/check/DellEMC_XtremIO_iSCSI/4bc8df5/DellEMC_XtremIO_iSCSI/440 : FAILURE in 2h 25m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"c500609d42783f0c4c614e606a0b90fb99713f81","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 19:30:23.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"feb12007569c02dd8a0bd2c3533700aa81998ca8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 20:35:09.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"753e8678c253f377f35a7dcf641278b1751efdca","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-06 21:00:05.000000000","message":"Patch Set 5:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/788363/5/fc : FAILURE in h m 0s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/788363/5/iscsi : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":5},{"id":"8dc4694de6744a27597618b184d42846096c694f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-09 13:33:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2e35a25b0be15734996412d91ea6fed1f29d3979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:50:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b46c0c720cf04b7093130fef4421045f : FAILURE in 7m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e214b1b1c0e64fcb92b5c3880b8d0530 : FAILURE in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37eb6a9f1524ee6a4ad7687252cc797cf283ad96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 13:56:43.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"0e5c70ae1f031a7939d68476300b84f48d018ede","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-09 14:02:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/be44c7e/ : FAILURE in 23m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/753a50e/ : FAILURE in 24m 45s","accounts_in_message":[],"_revision_number":6},{"id":"a9acf73519e77bf25f4eff62cc09d2d7f9080b30","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-09 14:11:26.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/6/check/DellEMC_Unity_FC/c1027c2/DellEMC_Unity_FC/404 : FAILURE in 36m 43s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/6/check/DellEMC_Unity_iSCSI/e68312b/DellEMC_Unity_iSCSI/380 : FAILURE in 29m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"00d354350a067efaf0a576f171ec812c18209fcf","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-09 14:13:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/6/check/DellEMC_XtremIO_FC/17fb4fb/DellEMC_XtremIO_FC/472 : FAILURE in 39m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/6/check/DellEMC_XtremIO_iSCSI/289f1d5/DellEMC_XtremIO_iSCSI/463 : FAILURE in 33m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"7fa1cc09414cb46b3c060a487861b2c4d0aa0c3d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 14:14:01.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"a246b051f54f8c091b8ffc2596898541755338fe","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-09 14:27:14.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/63/788363/6/check/DellEMC_PowerMAX_FC/6f30cd0/DellEMC_PowerMAX_FC/234 : FAILURE in 39m 38s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/63/788363/6/check/DellEMC_PowerMAX_iSCSI/ec100ac/DellEMC_PowerMAX_iSCSI/218 : FAILURE in 26m 06s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"3e5ba014ba683be62211a4ff5559f1f1298682ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 14:28:08.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"a8835a0ecea129e0a7d3d492d7aa4f27e371f62d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-09 14:40:22.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"84206c480fbc885fc94c0a8fdd8b148468daf258","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 14:40:47.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"70309f705d0e3bd48199bd488a4bc677468977e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 14:49:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78c4d2ef21ea49e3ba3bfd4488d96d48 : FAILURE in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eecbcd6c4cd24a60a3d80ad937c3a15a : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daca174ee2b04fd8947b8db72a7172f8 : FAILURE in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79d5a4ed0ff4aa2bac93118e996bf41 : FAILURE in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78af1bc0a8eb4e339d00cf8fb005e8fd : FAILURE in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609a0c92393d4ababa3f7d0744197c57 : FAILURE in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/4bbd127d5cde44c286169e04f53e28be : FAILURE in 1h 01m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f26a51a6b99d475cb8867c9e19bb83da : FAILURE in 27m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/16f0eec107f547c0b17c2df7b46f4017 : FAILURE in 6m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a1c2df545584fb2be0880adef263a96 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c82f6134e5a47dea6f6eb0c9c79d288 : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85dd4f34f5a24bd7ab6f38f9152e3959 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3af976728bb041cb9b99d9089ff538e6 : FAILURE in 23m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37565cb93c1a450dba0f9ca824d80e98 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/345b88f99bf04b84b6cfb5ca54eaf595 : FAILURE in 25m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0a69c09ad72b41848f2890b3a07f6ec1 : FAILURE in 34m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d0f200d99a46406884c676a0702fc998 : FAILURE in 27m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/94dee41743fc4b92b3c84bcc53f10010 : FAILURE in 1h 13m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d54a688cb66a4ae882eb6c529cce6a4a : FAILURE in 22m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c399e16fe414107b994f5d265225110 : FAILURE in 30m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b12200017f0423f91af62cde64c03e4 : FAILURE in 28m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7660a0fc2124e0a8e982513828dd2a4 : FAILURE in 30m 25s","accounts_in_message":[],"_revision_number":6},{"id":"654e37a7c4642723507240ee0cc4b4ecc66a637c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 14:58:25.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"cd6a0a438f946fe01935e407b1195e02d02fab39","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 15:31:26.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f1dd93c/f1dd93c.html : FAILURE in 52m 06s","accounts_in_message":[],"_revision_number":6},{"id":"6cc777c4b85d12a6b8b3d0525787f764ea7fcf94","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 15:35:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/6/check/iscsi-driver-primera/eaf9781/eaf9781.html : FAILURE in 53m 59s","accounts_in_message":[],"_revision_number":6},{"id":"5c219d43ba9e5db11c690f349ba6c517d813860a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-09 16:37:38.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 50m 32s","accounts_in_message":[],"_revision_number":6},{"id":"531d6abfa424fc98410258d4c9d45df040265f09","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-09 16:54:11.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-6/ns5_nfs : FAILURE in 1h 8m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-6/ns5_iscsi : FAILURE in 1h 9m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"17ca4dcd386fcd7db3479f340af97a693e43a598","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-09 19:52:20.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/6/iscsi/ : FAILURE in 4h 51m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/6/smb/ : FAILURE in 3h 08m 35s","accounts_in_message":[],"_revision_number":6},{"id":"e539ed011270277e0a0f79c6fe140719e530df5b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 20:14:07.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"f63110cd56db24ec63f1f905e9b4b053f7097f4e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-10 05:03:52.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/6/20210710124613/NEC-ISCSI : FAILURE in 1h 15m 59s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/6/20210710124613/NEC-FC : FAILURE in 1h 17m 37s","accounts_in_message":[],"_revision_number":6},{"id":"7fb6e8d636b06f002d6f9783629c3229864a05a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-10 05:36:40.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"10634564bbf15fbec881e877c384037766fa22e9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-12 05:57:25.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/788363/788363/6/check-not-vote/ext-cinder-zuul/a86bbcd : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":6},{"id":"80605b2df0fb2e0c4abb7ae24bdc43f0614fc07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-12 14:21:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a38a6846a864faac7d3410f8b2e5786b9e91ab99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:45:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcc8b1c425844aed8fed585efce452e1 : SUCCESS in 16m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b421a14e2604eb49e03950f198ea3cc : SUCCESS in 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ebf9f44a1825e0f642e4d535f1b60443ee09790d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-12 15:25:37.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/7/check/iscsi-driver-primera/f5518b8/f5518b8.html : FAILURE in 1h 02m 22s","accounts_in_message":[],"_revision_number":7},{"id":"323de358c93d09805a2ae99dd123e3854df9f306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:26:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d26d057850f499eb950d07f343ce93b : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e008ff07c547fe9a30d9f6d2b23006 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99184a138534bf1badcd103ff6d15b2 : SUCCESS in 9m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75b35ff1911b44758da9432b87409ef3 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2df14208b4d44bbe9c8f1a1e2d2d9275 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c774a1bcb134f19a4604de93dd76410 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/8b9e04a70c5c4000b07525def500e86a : SUCCESS in 1h 15m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb3d984a406d4a5dbe81b39ee392d5f9 : SUCCESS in 1h 01m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1cd66f7d0b746c0a7fe062034145865 : SUCCESS in 11m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/66e6086993c8485da65e78a0951734f7 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/377e965603114ef28d0d08b3490335e1 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dff3a7772941457680ccbfc1f9a5b7c5 : SUCCESS in 8m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c8b7544216ff41a898460f9e90343ed8 : FAILURE in 31m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eada596eb72e418694a07aba00c21193 : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/774a3c0c73dc48ac82d00cc5b7b218df : SUCCESS in 1h 06m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b0fd9490b56542eaa34725cbc19cf703 : SUCCESS in 1h 24m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a48fe316a8114ef9af10c54ddbbf72f7 : SUCCESS in 1h 19m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6100591603e84a3391844cacb9baa863 : SUCCESS in 1h 11m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/098efeafee404395a3a879f221651dca : SUCCESS in 1h 08m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/417e3e29be5944fbae2a13a6746380cf : SUCCESS in 1h 37m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d42221d5c60e4144952495e92b61273d : SUCCESS in 2h 02m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee13c82d14874d1c928c13c245c10a0d : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":7},{"id":"d23f027f8e6f161d7a355d64239d19742ac6b867","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-12 16:26:42.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/7/check/tempest-dsvm-full-focal-py3/1ad04dd/ : FAILURE in 1h 14m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"453e5b22eefdd282e69759f36ed7fd5213b2ecb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-12 16:27:39.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"0d9194cc286b0f6c212929274b15b1eafd56a84b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-12 16:29:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c7754c7/c7754c7.html : SUCCESS in 2h 06m 19s","accounts_in_message":[],"_revision_number":7},{"id":"e86290d56e4ce73411ac907f25078acb323a8c07","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-12 16:40:15.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/7/iscsi/ : FAILURE in 1h 20m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/7/smb/ : FAILURE in 56m 10s","accounts_in_message":[],"_revision_number":7},{"id":"8e0a801e7955313cc2cc90bf55a278d6c993647a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-12 16:40:38.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"e894b9799e111cc3fa0e5c32ca59422650c7721a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-12 16:44:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/63/788363/7/ : SUCCESS in 1h 53m 34s","accounts_in_message":[],"_revision_number":7},{"id":"8628c0dda2936e77e92931719de46671a521641b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-12 17:26:50.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cf80894/ : SUCCESS in 1h 14m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6d86ca4/ : TIMED_OUT in 3h 04m 04s","accounts_in_message":[],"_revision_number":7},{"id":"a858bc7696bfeabeb34ff2f6a2637319b7ff6bbb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-12 18:02:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/7/upstream-check/cinder-cDOT-iSCSI/bfeca86/ : SUCCESS in 1h 32m 33s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/7/upstream-check/cinder-cDOT-NFS/f3676e5/ : SUCCESS in 33m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/7/upstream-check/cinder-cDOT-FCP/342ee7c/ : SUCCESS in 55m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"56cb421f0c589949b9f83392d6ba3a3ebe5961df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-12 19:11:14.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"12d28d32061b4f89cc3735843e9f578ceca1dc47","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-12 19:25:57.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/7/upstream-solidfire-check/cinder-solidfire-iSCSI/42da7a7/ : FAILURE in 2h 55m 54s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"19d6fc715a8c23784fd50824dcb6619900456ca7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-12 19:33:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/7/check/DellEMC_XtremIO_FC/523e8da/DellEMC_XtremIO_FC/486 : FAILURE in 1h 24m 12s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/7/check/DellEMC_XtremIO_iSCSI/f727090/DellEMC_XtremIO_iSCSI/476 : FAILURE in 1h 55m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"e1672b89279953a5ba466dc7ec0f7ab5aee757a4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-12 19:38:06.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/7/check/DellEMC_Unity_FC/783992a/DellEMC_Unity_FC/418 : FAILURE in 1h 20m 33s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/7/check/DellEMC_Unity_iSCSI/7616afb/DellEMC_Unity_iSCSI/393 : FAILURE in 1h 59m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"2dc0e5e064afd8e3095df8b463c2ed63ca18941a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-12 20:00:37.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/7/check/DellEMC_PowerStore_iSCSI/55be358/DellEMC_PowerStore_iSCSI/236 : FAILURE in 1h 10m 55s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/7/check/DellEMC_PowerStore_FC/2bbdbf0/DellEMC_PowerStore_FC/237 : FAILURE in 55m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"889628ac89fc40f82d128d0b4f8406f2d66ca212","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-12 20:24:06.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-7/ns5_nfs : SUCCESS in 4h 38m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-7/ns5_iscsi : SUCCESS in 4h 46m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"06173c71831b7b354ac0a3bd3813ab0e4e657b50","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-12 20:39:20.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-7/PVME-iSCSI : FAILURE in 9:25.95\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-7/PVME-FC : SUCCESS in 48:05.83","accounts_in_message":[],"_revision_number":7},{"id":"9d761db204190d7c3dc2976e1ead620e6e85f11b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-12 21:56:27.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"1124aa107c737610b9e35cbb7093db73cce1ad9a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-12 22:14:28.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-7/Seagate-iSCSI : SUCCESS in 41:51.94\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-7/Seagate-FC : SUCCESS in 38:54.92","accounts_in_message":[],"_revision_number":7},{"id":"69a4ec877a832d5d394a3e77c3a309715869b7af","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-12 23:07:13.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-7/HPMSA-iSCSI : FAILURE in 1:00:33\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-7/HPMSA-FC : SUCCESS in 31:33.22","accounts_in_message":[],"_revision_number":7},{"id":"36e22f7d3b5c04f17580625b2bc67961e3aaaf0d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-12 23:12:08.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-7/Lenovo-iSCSI : SUCCESS in 48:51.67\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-7/Lenovo-FC : SUCCESS in 44:48.53","accounts_in_message":[],"_revision_number":7},{"id":"b5c4dc686f2d9c33672a84ebd24ab1081cc082ad","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-13 02:11:59.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/63/788363/7/check/DellEMC_PowerMAX_FC/70e2193/DellEMC_PowerMAX_FC/246 : FAILURE in 5h 23m 09s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/63/788363/7/check/DellEMC_PowerMAX_iSCSI/845002b/DellEMC_PowerMAX_iSCSI/230 : FAILURE in 5h 18m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"834877a4846146ed4b6769e4607dbda28e243c31","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 03:55:23.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"0e8fba554a5440f32429ed0c681b781044053cf9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 02:40:46.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/7/20210715085900/NEC-ISCSI : SUCCESS in 2h 31m 41s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/7/20210715085900/NEC-FC : SUCCESS in 2h 41m 42s","accounts_in_message":[],"_revision_number":7},{"id":"2f88c858cc3adf2697660ecc16f344fd59b462c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-15 17:35:05.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"6a52d458b1ac712519addd6cfc78873f5aab1333","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 17:36:51.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"6ddae43eeec87a447ea395c276be3ffdfefe70c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:54:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/221d21ad2f274ee1bd996a54c62d2f6b : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d156125c0f94d93b9ee32fb45606aaa : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"01348afa226ce0ac5a59060e3c2a6a5b9cf47871","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-15 18:25:24.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"2f3780f6f53a8fc04c2abce5774161c4a0e579eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 18:28:07.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"1f258ace87020436c813b941965a9befedd1eaf8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 18:41:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/8/check/iscsi-driver-primera/cb85f6b/cb85f6b.html : FAILURE in 1h 04m 05s","accounts_in_message":[],"_revision_number":8},{"id":"3c89621618af02fb21125518ee9bd037f74c11d4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-15 19:07:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d7158cd/ : SUCCESS in 1h 31m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a93910a/ : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":8},{"id":"68e80ce56e8db3cf1bdbb31cfd4ded50b914e12f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-15 19:17:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/8/iscsi/ : SUCCESS in 1h 07m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/8/smb/ : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":8},{"id":"a953317da3153f7725f9f2bde6bc90ea3cb7ea46","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-15 19:21:45.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/63/788363/8/ : FAILURE in 46m 26s","accounts_in_message":[],"_revision_number":8},{"id":"effa7290e80aecb0bcca1fd705284eacaf9aea45","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-15 19:29:32.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788363/788363/8/check-not-vote/ext-cinder-zuul/b1cc948 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":8},{"id":"2e806a88d01910e85b726bca9ce61921bbd3972d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:35:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9251b12f194e4b6f8e6242068f001652 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c7fb5307eb48c98393593fd6aca222 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d96db03dcc42cdbe81ec692edb591b : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8463c74246d43589e993bb5dcd0235f : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d55d0795394e5eaf84c0f7b0c3ebad : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/cf81dd9108404548b8e9eee0e6b27795 : SUCCESS in 1h 09m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/787c693759cd453f8a99d01235f6bd07 : SUCCESS in 1h 14m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/49159f6831c94ce8be4eed2c40179e77 : SUCCESS in 14m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86da4ea1213d4c58b087ff04f45f8263 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d5514d877c74bb7830a4f422587f5ac : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cd3078363004f8fb99b5cbf9eea5f10 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2fda15bd4d34b8f93d01deae5078b12 : FAILURE in 27m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0347c9b590844348ccd12b168ff07fa : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a644d92a51654424863454511220aff9 : SUCCESS in 1h 06m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1fff8349ac0248618bd4af180fa46d99 : SUCCESS in 1h 27m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/da98a478f0104d6787601d0b5cc5b752 : SUCCESS in 1h 45m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bcfb31d5c1c24c97987d1f2a7beea493 : SUCCESS in 1h 04m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bef495b3abb74307a26347b7242dc764 : SUCCESS in 1h 36m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/45f6e2fe9116487b8c29a37dbaf782e9 : SUCCESS in 1h 06m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b089bbf696a4149aeae81ae0430eb5f : SUCCESS in 1h 58m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/805e2faadfb243569645efce7edeb3d2 : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":8},{"id":"3ff64742f85f5eccc0bea66b80c44c9cc87f9d12","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 19:51:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/01d502a/01d502a.html : SUCCESS in 2h 14m 16s","accounts_in_message":[],"_revision_number":8},{"id":"a14738fbed3ece7f35bf79d89e5f2a4397e4309d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-15 20:23:48.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/8/check/DellEMC_PowerStore_iSCSI/185b628/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/8/check/DellEMC_PowerStore_FC/08e58c3/DellEMC_PowerStore_FC/245 : SUCCESS in 46m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"f4b8b5b4df56c9c62e73f9fdbb9b142733006ef8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-15 20:35:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/8/upstream-check/cinder-cDOT-iSCSI/545c782/ : SUCCESS in 43m 07s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/8/upstream-check/cinder-cDOT-NFS/22bced9/ : SUCCESS in 44m 07s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/8/upstream-check/cinder-cDOT-FCP/b0e4826/ : SUCCESS in 57m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"95c3b23e97ff6a79a049f90724fbdfed22d6fe66","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 20:46:46.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"7e187d3e04b8f56fc277500871ca20d57d9509bf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-15 21:03:19.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/8/check/DellEMC_Unity_FC/f3f3c0c/DellEMC_Unity_FC/470 : FAILURE in 1h 17m 21s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/8/check/DellEMC_Unity_iSCSI/deff1c2/DellEMC_Unity_iSCSI/441 : FAILURE in 1h 14m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"333271a6b0cb1f94fe259d1400db6d8f4d8e97ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 21:12:35.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"634f4c3771a5a123aff667dfd0569794639f087e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-15 21:29:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/8/upstream-solidfire-check/cinder-solidfire-iSCSI/71f3145/ : FAILURE in 1h 51m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"920709f86fbcbbfc102790441638b8b769154ae5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 22:46:32.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/63/788363/8/check/DellEMC_XtremIO_FC/4ce5aff/DellEMC_XtremIO_FC/534 : SUCCESS in 1h 16m 20s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/63/788363/8/check/DellEMC_XtremIO_iSCSI/a021d21/DellEMC_XtremIO_iSCSI/520 : FAILURE in 2h 00m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"b2c8fb5a695e23a085de38f6c2051677ce9f0880","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-15 23:16:14.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-8/ns5_nfs : SUCCESS in 4h 51m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-8/ns5_iscsi : SUCCESS in 4h 57m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"6c15dc0ccd3b1ba42ad01fb6db2e03d89c01114f","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-16 00:58:00.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-8/Seagate-iSCSI : SUCCESS in 43:35.41\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-8/Seagate-FC : SUCCESS in 40:08.68","accounts_in_message":[],"_revision_number":8},{"id":"0e9e2dd7802e5f908e2afa0af9d1c2ca3f063729","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-16 01:23:03.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-8/HPMSA-iSCSI : FAILURE in 59:04.21\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-8/HPMSA-FC : SUCCESS in 35:52.05","accounts_in_message":[],"_revision_number":8},{"id":"4d3e1c3dd685fa9a8b8018d0e22c371539938090","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-16 08:19:23.000000000","message":"Patch Set 8:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-8/PVME-iSCSI : FAILURE in 3:04:04\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-8/PVME-FC : SUCCESS in 46:45.60","accounts_in_message":[],"_revision_number":8},{"id":"7ea632126d4dec9511a418e60e16d21708d2ade4","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-16 17:42:47.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-8/Lenovo-iSCSI : SUCCESS in 50:42.31\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-8/Lenovo-FC : SUCCESS in 49:57.82","accounts_in_message":[],"_revision_number":8},{"id":"7fe1e76c9264c0892fe77d236956d9d643bf9d5e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-17 06:29:56.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/63/788363/8/check/DellEMC_PowerMAX_FC/4f5868c/DellEMC_PowerMAX_FC/264 : FAILURE in 5h 59m 24s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/63/788363/8/check/DellEMC_PowerMAX_iSCSI/a6de229/DellEMC_PowerMAX_iSCSI/250 : FAILURE in 8h 24m 00s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"9f41af1135b63ff58c1b48ada81aa0a047a04b40","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-17 07:13:41.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"56d4dc80038f704fd2bc99e64f32e3075b85f5f8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-18 12:17:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/8/20210718183604/NEC-ISCSI : SUCCESS in 2h 25m 46s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/8/20210718183604/NEC-FC : SUCCESS in 2h 41m 23s","accounts_in_message":[],"_revision_number":8},{"id":"7b7cd793e15a29f187c31c8c21e585528e10d73e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-28 16:01:03.000000000","message":"Patch Set 7:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/788363/7/fc : FAILURE in 0h 55m 29s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/788363/7/iscsi : FAILURE in 1h 34m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c960eff1435a2b665e21b25f229e6fbb7813f361","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-28 16:01:07.000000000","message":"Patch Set 8:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/788363/8/fc : SUCCESS in 1h 06m 27s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/788363/8/iscsi : SUCCESS in 0h 59m 16s","accounts_in_message":[],"_revision_number":8},{"id":"19349ee1c97605749e96c725dcca3c360b0cf8e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-09 13:09:18.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"b42a75d732ebf050da3da6bb6f412a33c14bb7e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 13:32:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/218e0e3540244011bf229c1747daffce : SUCCESS in 15m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb58b9a942be4a2993457f8d678ab7b3 : SUCCESS in 21m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ce3e4fa10b8eb69aeab040156b1d95eb4512a974","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-09 15:08:37.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/9/iscsi/ : SUCCESS in 1h 08m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/9/smb/ : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":9},{"id":"9e102b58a584a6812dc188d3f5c5f4e8c0cb13d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:23:31.000000000","message":"Patch Set 9: Verified+1\n\n(2 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664557372e694668880908a07b7596ad : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5eeaf7821474a5d851be0d30601c150 : SUCCESS in 12m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa15dff59b404b5382dc7417e5b6a824 : SUCCESS in 10m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07da40f7ae964ffd93beda8f72a73cb9 : FAILURE in 14m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108e1eeaf07a404d8bcecb30b0d1c8dc : SUCCESS in 17m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/cb21447163f94b09b9651732ee7a2e86 : SUCCESS in 1h 02m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84625454a93a4329bcd60fddb1f7a4a0 : SUCCESS in 1h 06m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/acc6f3aab24d4a9f978c826ff904133a : SUCCESS in 14m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6cb356d6c5a4f93a329a517368107df : FAILURE in 6m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30576e74b18e4fa4b8f085d666de298d : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e76e3f040d824818b9200c7d34bfcb73 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e3533e9a5d24303931c28869a931c36 : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5237706f6b7b406a92c22f93e9d584bf : SUCCESS in 38m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7409350f7c54d7aaf4327e3578470f1 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4af9b77278c044d3a1c5ada33287dfdc : SUCCESS in 1h 26m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8dd64995a87d4e25b8b8578d9ede8fdf : SUCCESS in 1h 21m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0ea9747eb3e4061972acfcbe4b932cc : SUCCESS in 1h 19m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e114c175a5b34dfe98b53fb561e821f2 : SUCCESS in 1h 11m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/05f43555f6164b649cdfd8d7a7a02e83 : SUCCESS in 1h 10m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c27a2104e98a45ec926827ac748e6a64 : SUCCESS in 1h 16m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07243e8265a74aababd597976331d65c : SUCCESS in 2h 09m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdb46eab2e0c411da79935bd7078e3c3 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":9},{"id":"bb5e83dfee94c84f516b272b9b15d9ec73ea97fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-09 15:39:29.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"e56517f0c689e34a6075a82429cb661c974ab1c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:03:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/604a06b7e7a34783a3454c5dc12b491d : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ac4cc0a5dae4f13904aa2a609e49935 : SUCCESS in 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1ef846f02d4504c03754e591585887004fb1b3e8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-09 18:01:56.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/10/check/iscsi-driver-primera/0ec6b37/0ec6b37.html : FAILURE in 1h 04m 59s","accounts_in_message":[],"_revision_number":10},{"id":"c544ea30d38cac3658f327afab00b244e74ceeed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 18:05:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8d72fd543a4fa6a12f27c3f09681fb : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b96b3c7e58f4ac4ac02415c715e2ae5 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/190a86e16a9c4c789ec34c60de7aca27 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae7074ab188e4b3ebf54db9ad0f5b99f : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72989938be74a0cbc0ba3ba0b9c1555 : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b10fa529555a46bebf9d2b792f1f4a13 : SUCCESS in 1h 01m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a6df121c4e34f9aa40a74340397e5c4 : SUCCESS in 1h 08m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4a295cedc7614cdb836f25a05f70e364 : SUCCESS in 12m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48581d21afc74870bad499579985158d : SUCCESS in 5m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed9fdf4b74a444239f1824d29fe56a7f : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a09c7565b9a4225ac52c81302f67f32 : SUCCESS in 7m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1fb4b14d5ac04b4bb8578e8fd845122d : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e29d2b162204487f9556a484abb89716 : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5734765f7ff24ce5806ec93c4424accc : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d30e1c4b4a443959d817eff796f4f3e : SUCCESS in 1h 23m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9e91f23deb2040c782e7cc76726b5964 : SUCCESS in 1h 29m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3168e8195a0d4b6fbb0e60f1db125853 : SUCCESS in 1h 25m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f1508f9d0fc7476dbc37a456a40534f2 : SUCCESS in 1h 13m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5ef66576fd2425fa09b2649e37793a1 : SUCCESS in 1h 07m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99c75332fbe7465884bd4f48d05d408d : SUCCESS in 1h 12m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91360b2b60f94676ae35fab770d19687 : SUCCESS in 2h 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faae2b2aabe14321965b04a0a922b9a6 : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":10},{"id":"6cb70c43a7b1bbe378e370ac8ea7cc3637b5d327","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-09 18:11:29.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/10/iscsi/ : SUCCESS in 1h 19m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/10/smb/ : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":10},{"id":"07a3b2b72b74e9c8c711ff0d4485e8e60b6dc972","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-09 18:21:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/9/upstream-check/cinder-cDOT-iSCSI/e601a25/ : SUCCESS in 37m 11s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/9/upstream-check/cinder-cDOT-NFS/2bca330/ : SUCCESS in 34m 54s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/9/upstream-check/cinder-cDOT-FCP/f35238a/ : FAILURE in 2h 53m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"60d40acab6802f25a322afdb08393408219e7e42","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-09 19:00:14.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/10/check/tempest-dsvm-full-focal-py3/ffd8c09/ : FAILURE in 2h 14m 41s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"adb21f331f36947ce36cd0a955e256d833782dd4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-09 19:37:19.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aa2057d/ : SUCCESS in 1h 17m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/90c3e87/ : SUCCESS in 1h 26m 52s","accounts_in_message":[],"_revision_number":10},{"id":"ea7ba87abdedc372596ec9854535907e118033f6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-09 19:40:17.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-9/ns5_nfs : FAILURE in 1h 17m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-9/ns5_iscsi : FAILURE in 1h 17m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"b42ec195060efd69a2503f462f27d4f8f7e45047","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-09 19:48:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/10/upstream-solidfire-check/cinder-solidfire-iSCSI/b8d0465/ : FAILURE in 1h 42m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"88a78f86548b2ebe235530ea34ebccd1e2a7d8a9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-09 19:59:56.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/10/check/cinder-storpool-tempest/f68ffe5/ : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":10},{"id":"e9d61c4fe9495b5875d2d3741eefba5f052a96f1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-09 21:00:40.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/63/788363/10/check/dsvm-tempest-cisco-zonemanager-job/bba5a4f/ : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":10},{"id":"8f2550f998ed511c4bddb6f79d334faa7bd8cbbf","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-09 21:06:16.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/10/upstream-check/cinder-cDOT-iSCSI/138a602/ : SUCCESS in 39m 16s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/10/upstream-check/cinder-cDOT-NFS/7c6319c/ : SUCCESS in 39m 07s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/10/upstream-check/cinder-cDOT-FCP/5aed71c/ : FAILURE in 2h 53m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"2cf53f8c8da0a7382336fcf6d579a95833c4a77d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-09 22:28:37.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)\n\nQuestion inline, otherwise I think this looks fine.","accounts_in_message":[],"_revision_number":10},{"id":"b3adb9ed79ff6472239c1710df0755f3865ed805","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-10 00:32:58.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/88fd98a/88fd98a.html : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":10},{"id":"a167b602f804ec0c213b8b4163a1d3cdadb541ec","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 01:01:30.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-10/ns5_nfs : FAILURE in 1h 5m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-10/ns5_iscsi : FAILURE in 1h 8m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"0d3d4f0e341d3c2c4e5354527a77bc32dea7e16b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 01:50:38.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/10/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":10},{"id":"b1ba0a701c3f7d7b4526a0aa1af62ac108ddfd58","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-10 02:39:09.000000000","message":"Patch Set 9:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-9/PVME-iSCSI : FAILURE in 14:54.98\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-9/PVME-FC : SUCCESS in 46:49.93","accounts_in_message":[],"_revision_number":9},{"id":"c82fbbcbae2d8b7e9e3bbd0012491e2899b3c583","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-10 06:45:59.000000000","message":"Patch Set 9:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-9/Seagate-iSCSI : SUCCESS in 42:12.09\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-9/Seagate-FC : SUCCESS in 37:00.82","accounts_in_message":[],"_revision_number":9},{"id":"9c3c96fb955cf2ad08e998079283d19fdd8bedd3","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-08-10 06:58:40.000000000","message":"Patch Set 10:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":10},{"id":"3f9a9bec170361fe2f87083f2392f4f7945cfd8b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 07:32:47.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788363/788363/10/check-not-vote/ext-cinder-zuul/fa3725c : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":10},{"id":"067a888ecf8d59c5dbe9f02afa4a09f201544355","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-10 09:17:07.000000000","message":"Patch Set 9:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-9/HPMSA-iSCSI : FAILURE in 59:39.08\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-9/HPMSA-FC : SUCCESS in 40:00.37","accounts_in_message":[],"_revision_number":9},{"id":"10baabcb9a96f231cd4dd9ab863f074e972ce6c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 09:36:08.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/10/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":10},{"id":"6487d53eadfd9a91b5b3ce50065512c93759c60c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-10 10:04:03.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/63/788363/10/check/DellEMC_SC_FC/8fe49b9/DellEMC_SC_FC/270 : FAILURE in 50m 14s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/63/788363/10/check/DellEMC_SC_iSCSI/b97849d/DellEMC_SC_iSCSI/266 : FAILURE in 1h 54m 41s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"5d02983ef179d11491f1214683f24ae7769bc400","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-10 12:31:58.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/63/788363/10/check/DellEMC_Unity_FC/a22aec6/DellEMC_Unity_FC/612 : SUCCESS in 1h 15m 10s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/63/788363/10/check/DellEMC_Unity_iSCSI/80d0873/DellEMC_Unity_iSCSI/578 : ABORTED in 3h 01m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"9e7fcbd5397fcaa5360ca9b396940294012ffe3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 12:34:03.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/10/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":10},{"id":"417e6e60a06de592755c32e0e9c7c9c514e42354","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 13:19:54.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"25ae418ec8f778c5a415bcdbb1ce2d6a07a5aff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 13:24:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4df315e388980d8ce8ff3aac1cdd33550ef65b2d","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-10 13:41:23.000000000","message":"Patch Set 10:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-10/PVME-iSCSI : FAILURE in 14:59.13\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-10/PVME-FC : SUCCESS in 46:22.72","accounts_in_message":[],"_revision_number":10},{"id":"7966c685da1d9fa732219ee4ccef8341a5761799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:55:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f45d140ca4144d7a7534e46fef8b1a3 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/572703459f164972a910d919ab6e63f3 : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8c8eda36b133f45438507b985281527929eec336","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-10 14:20:12.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/11/iscsi/ : FAILURE in 53m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/11/smb/ : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":11},{"id":"66564ca4406b9fde8ccf8f474fe071ca323d62a8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-10 14:26:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/11/check/cinder-storpool-tempest/372b7e7/ : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":11},{"id":"ec04a0dc2291aeba88f65922387595d7f9e9e9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-10 14:27:34.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"d123e7a3647ba26808967295ba4fb65c885256ab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-10 14:28:40.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"11ae7b479e6d61794385137af391a0dde7c233a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-10 14:35:13.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"6404ed0bd2211da41a3b75544ae2e6c84db1672b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 14:35:27.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"bb5954744920dedff1ba626297fe82dd8ad64032","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-10 14:35:29.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"6d761ec2a74c95e00eeea65f19f029895f779f36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:54:05.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b17249a64e704788a555753b490ffca4 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95fb778249b648038c4ca512b64d0e59 : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"64bb37b349e8552b84fdd24b175b89003f45c121","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-10 15:25:15.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788363/12/iscsi/ : SUCCESS in 55m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/788363/12/smb/ : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":12},{"id":"deae6652dc32082e2ae94c601da7753a83e907e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:38:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cead8ef4821140fe826f092ee7e53e5b : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64d3cea2ac9470d8c1e427ceedb50bc : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9110ae3985a04a99b9e82703d471be10 : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d73ae8765973415ca93395d997b0a3dc : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6890cc5238420b8a98b7e7fe90d8c5 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ab86c9f610c6424c8d945a7b89568627 : SUCCESS in 56m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5087815c847d4d759c0af46fea9818b4 : SUCCESS in 1h 07m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a161bf06661a4bfd88f07d8658db4bcd : SUCCESS in 11m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd2ee255536b4b7fa265bbdee8cbb3ec : SUCCESS in 6m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6303ba98ed864b72a41a5ff9709343c7 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e22b0a4f4c5b4f859eb39a2399eb1c15 : SUCCESS in 7m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ef25c8bbead445d096ff831615ce8df1 : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/554fff70d3a444e187836e51741766d6 : SUCCESS in 39m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e72133d440b402fbc6dafd7cf94e2ce : SUCCESS in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5931efb8374d470b89fc87db77e62b34 : SUCCESS in 1h 17m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28aaf21aac704b448305ffd3fcd7e6af : SUCCESS in 1h 38m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fdd8cef5349466e88be7d6cbc895ee0 : SUCCESS in 1h 24m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c23e6895c044e5390ec1398b60058d1 : SUCCESS in 1h 19m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a1ab8b67c25b45abad17bd121433ea0f : SUCCESS in 1h 08m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e5304a39c2642f8a9e1682f82de90af : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00b8bee3e4064e079661f766240a58af : SUCCESS in 2h 07m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc39ae458d44ed7bac9cbb1e68ed6fd : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":12},{"id":"9229d33f66133734e80e46bd4d28efa3cd992e12","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 16:42:39.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-11/ns5_nfs : FAILURE in 1h 17m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-11/ns5_iscsi : FAILURE in 1h 19m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"ba813e7acba3bfef83ddb8e421cebdc1b3a35201","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-10 16:53:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/63/788363/12/check/cinder-storpool-tempest/aa097dc/ : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":12},{"id":"78fd79295a2fc5e7d84ee618892918d25e7abe86","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-10 16:57:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/63/788363/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fcba332/ : SUCCESS in 1h 22m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/63/788363/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/52249ba/ : SUCCESS in 1h 28m 10s","accounts_in_message":[],"_revision_number":12},{"id":"cb314449ae87ed1f803d8c3d52e80b1d956b3a54","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-10 17:05:57.000000000","message":"Patch Set 9:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-9/Lenovo-iSCSI : FAILURE in 2:43:34\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-9/Lenovo-FC : SUCCESS in 50:39.53","accounts_in_message":[],"_revision_number":9},{"id":"9d461f41e25c6e186c4f5318de397fca2494723a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-10 19:01:12.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/63/788363/12/check/iscsi-driver-primera/e69b8ca/e69b8ca.html : FAILURE in 1h 09m 20s","accounts_in_message":[],"_revision_number":12},{"id":"3d1f17e0cc252aa16929e6e35293da5ea4630e98","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-10 19:26:19.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/63/788363/12/upstream-solidfire-check/cinder-solidfire-iSCSI/b3c7c24/ : FAILURE in 2h 46m 35s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"e8b9ae912063f82568f597626a1924b22b017074","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-10 19:37:24.000000000","message":"Patch Set 10:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-10/Seagate-iSCSI : SUCCESS in 40:05.47\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-10/Seagate-FC : SUCCESS in 36:47.04","accounts_in_message":[],"_revision_number":10},{"id":"977f96e8f93d1933bf1d898e0cc89bf7fca7737f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-10 19:38:57.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/63/788363/12/upstream-check/cinder-cDOT-iSCSI/558cad5/ : FAILURE in 2h 59m 38s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/63/788363/12/upstream-check/cinder-cDOT-NFS/13e4ae5/ : FAILURE in 2h 47m 43s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/63/788363/12/upstream-check/cinder-cDOT-FCP/d780587/ : FAILURE in 1h 59m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"743a3627c3df3ce54ae960a70cfe1ccaa09a86d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 20:18:36.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/11/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"6a9e4740f3ac11f68d6c93f1b4b9cbb45d479ba1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-10 21:16:40.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":12},{"id":"0aecc28289e80e1ae3adffb65c76a689d1b1b2d4","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-10 22:39:36.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nLGTM, thanks for the detailed reply.","accounts_in_message":[],"_revision_number":12},{"id":"2c25a02a336b85fddc55c09db2e95905b0f9c73f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 23:22:41.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-12/ns5_nfs : SUCCESS in 4h 5m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-63-788363-12/ns5_iscsi : SUCCESS in 4h 7m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"2cdefe641021bc10522790fa0f6053de8a09ab6f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-11 00:02:35.000000000","message":"Patch Set 10:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-10/HPMSA-iSCSI : FAILURE in 59:51.19\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-10/HPMSA-FC : SUCCESS in 34:16.04","accounts_in_message":[],"_revision_number":10},{"id":"aa146a9a425b5a0442009575431d2a0f981414b0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 01:14:57.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/63/788363/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/41ae31b/41ae31b.html : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":12},{"id":"ceaeeb42f969af87f0fef2e2e53d6740ed6a6cdc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-11 04:43:25.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/63/788363/12/check/iscsi-driver-nimble/fbcfc1e/fbcfc1e.html : FAILURE in 2h 05m 02s","accounts_in_message":[],"_revision_number":12},{"id":"fc05283c9b008f6f52aca51a47ea98dd2ff1d0f8","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-11 08:55:55.000000000","message":"Patch Set 12:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788363-12/PVME-iSCSI : FAILURE in 14:41.80\n* PVME-FC http://os-logs.tristero.net/dell-ci/788363-12/PVME-FC : SUCCESS in 46:17.00","accounts_in_message":[],"_revision_number":12},{"id":"3e093390a8eb9998b0e5da1c4e4b756f9e4dc585","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-11 12:11:25.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/63/788363/12/check/DellEMC_PowerStore_iSCSI/7af750b/DellEMC_PowerStore_iSCSI/326 : ABORTED in 3h 00m 05s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/63/788363/12/check/DellEMC_PowerStore_FC/0851c0e/DellEMC_PowerStore_FC/363 : SUCCESS in 50m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"89f0e28eeceb4fc76f060e8fe9261ee63642336c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-11 18:01:35.000000000","message":"Patch Set 12:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788363-12/Seagate-iSCSI : SUCCESS in 43:12.31\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788363-12/Seagate-FC : SUCCESS in 41:45.04","accounts_in_message":[],"_revision_number":12},{"id":"1cd5fcb6f89988366f0e125bae8d7f4d3ec83103","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-11 19:04:28.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/12/check/tempest-dsvm-full-focal-py3/e40a432/ : FAILURE in 2h 28m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/63/788363/12/check/devstack-plugin-ceph-dsvm-focal-py3/744aaf1/ : FAILURE in 51m 47s","accounts_in_message":[],"_revision_number":12},{"id":"ee8613edfd68f90edd52a0a97bdeae5ee0a8cc07","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-11 21:12:12.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788363-10/Lenovo-iSCSI : FAILURE in 3:49:01\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788363-10/Lenovo-FC : SUCCESS in 2:15:21","accounts_in_message":[],"_revision_number":10},{"id":"47583a072d869fbad98612afad8d32008c9aa098","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-12 00:49:42.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788363-12/HPMSA-iSCSI : FAILURE in 1:02:44\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788363-12/HPMSA-FC : SUCCESS in 36:21.61","accounts_in_message":[],"_revision_number":12},{"id":"6e99e0ce2f225132ba7be355fee6bfdf4d59f626","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-12 02:48:29.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/63/788363/12/check/DellEMC_PowerFlex/86efcc8/DellEMC_PowerFlex/381 : ABORTED in 3h 00m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"5f777f55f0337d49b3f472d01c417d292368149d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-12 17:01:39.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/63/788363/12/check/dsvm-tempest-TOYOU-ACS5000-driver/76d24c4 : FAILURE in 1h 29m 23s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/63/788363/12/check/fc-test/b9d98a4 : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":12},{"id":"9aecf118befda0491e13cb49af64ef6a17afae54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-12 18:19:56.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/63/788363/12/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"44a57b814c242ec0f9a073b15931ca1169e9049e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-08-13 08:35:12.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/12/20210813145436/NEC-ISCSI : SUCCESS in 2h 39m 50s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/63/788363/12/20210813145436/NEC-FC : SUCCESS in 2h 40m 31s","accounts_in_message":[],"_revision_number":12},{"id":"f81fd934f1a028213a01920febd7346147acf1e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 18:51:17.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"ec6e31c0f9dab375d5249932597df3beb4aa26e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 22:29:41.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd35aabdd54b4a7087dafc14d4789a54 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1562b5259f444a4899cba6a1bb698b10 : SUCCESS in 9m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03a633f9cc814709b0a80ad643daa1a2 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24da3922b12b4c9792e4672d0e7ae631 : SUCCESS in 12m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/6c239284fbb0414d9df1b638b944944d : SUCCESS in 58m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dbf3e2782cde40e5bf28ab398cd0cc6b : SUCCESS in 1h 04m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/52281274f33e4e20aa6d9ebe3c613635 : SUCCESS in 1h 08m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5709923aa9d41b393ca261776481227 : SUCCESS in 1h 08m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b34d6c150dd4934a6bfec52df0a3909 : SUCCESS in 39m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2aa002023d184d8082d642aa533a9139 : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":12},{"id":"78d4b15e51130099b53fc6e47828e5a4d5316cf6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 22:29:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"15021d27be12f02c0a9ef3229f1849bdccfa120b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-08-21 08:31:42.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/63/788363/12/ : SUCCESS in 1h 53m 52s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"8c46c09ad5d4ae6b87400b678a2bd7afc4af6631","revisions":{"f2c6fa08497e99e3b96aa4f2c72b3c3be807cc41":{"kind":"REWORK","_number":1,"created":"2021-04-27 21:46:11.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/1"}}},"commit":{"parents":[{"commit":"b973743d4a24b0464a88bdce54c119419709e69d","subject":"mypy: backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b973743d4a24b0464a88bdce54c119419709e69d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 21:32:32.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2c6fa08497e99e3b96aa4f2c72b3c3be807cc41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2c6fa08497e99e3b96aa4f2c72b3c3be807cc41"}]},"branch":"refs/heads/master"},"d7593ca44679626407fbbfb278880913c03c5fbd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-30 12:26:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/2"}}},"commit":{"parents":[{"commit":"5dcc1362aba5daef0a2904046ee093132c954213","subject":"mypy: backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dcc1362aba5daef0a2904046ee093132c954213"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-30 12:17:59.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7593ca44679626407fbbfb278880913c03c5fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7593ca44679626407fbbfb278880913c03c5fbd"}]},"branch":"refs/heads/master"},"3ae9f172ef5f56bd5924579b94beff665b2d7113":{"kind":"REWORK","_number":3,"created":"2021-05-10 18:28:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/3"}}},"commit":{"parents":[{"commit":"9c40d92dbd7233b092142b63860548980defe2f1","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c40d92dbd7233b092142b63860548980defe2f1"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-10 18:25:51.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ae9f172ef5f56bd5924579b94beff665b2d7113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ae9f172ef5f56bd5924579b94beff665b2d7113"}]},"branch":"refs/heads/master"},"fc069a11f98c29b68c101c02ff3f99c5eed54ecf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-23 15:25:48.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/4"}}},"commit":{"parents":[{"commit":"159c5a417bf614226432604fbd5ecad7f6923c37","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/159c5a417bf614226432604fbd5ecad7f6923c37"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-23 15:25:48.000000000","tz":0},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc069a11f98c29b68c101c02ff3f99c5eed54ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc069a11f98c29b68c101c02ff3f99c5eed54ecf"}]},"branch":"refs/heads/master","description":"Rebase"},"fba2b954cfcfdd99249df3b62ecd8f7b01e54312":{"kind":"REWORK","_number":5,"created":"2021-06-30 20:45:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/5"}}},"commit":{"parents":[{"commit":"72c9ef8b8b7ca544b9398f0044edc684cbda1c64","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72c9ef8b8b7ca544b9398f0044edc684cbda1c64"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-30 20:44:49.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fba2b954cfcfdd99249df3b62ecd8f7b01e54312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fba2b954cfcfdd99249df3b62ecd8f7b01e54312"}]},"branch":"refs/heads/master"},"53ca929dfea133eea58542e989c67c5c9c1d71ea":{"kind":"REWORK","_number":6,"created":"2021-07-09 13:33:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/6"}}},"commit":{"parents":[{"commit":"72c9ef8b8b7ca544b9398f0044edc684cbda1c64","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72c9ef8b8b7ca544b9398f0044edc684cbda1c64"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-09 13:33:10.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53ca929dfea133eea58542e989c67c5c9c1d71ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53ca929dfea133eea58542e989c67c5c9c1d71ea"}]},"branch":"refs/heads/master"},"16bdba196c8f43215d01b8e5783a80c602d73dd9":{"kind":"REWORK","_number":7,"created":"2021-07-12 14:21:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/7"}}},"commit":{"parents":[{"commit":"914750a55dceebaf1ec6c750f8a616d031ee922f","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/914750a55dceebaf1ec6c750f8a616d031ee922f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-12 14:20:41.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16bdba196c8f43215d01b8e5783a80c602d73dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16bdba196c8f43215d01b8e5783a80c602d73dd9"}]},"branch":"refs/heads/master"},"057f06e857e420fb480ec91da3af51cea7f77177":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-15 17:35:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/8"}}},"commit":{"parents":[{"commit":"99157cfcd9208b52a961ef545603e538b81ab4f9","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99157cfcd9208b52a961ef545603e538b81ab4f9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-15 17:35:05.000000000","tz":0},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/057f06e857e420fb480ec91da3af51cea7f77177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/057f06e857e420fb480ec91da3af51cea7f77177"}]},"branch":"refs/heads/master","description":"Rebase"},"1aaaaff35ab3f7faeedcc3180f0f167c8fb53e5a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-08-09 13:09:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/9"}}},"commit":{"parents":[{"commit":"a509b0a8ccb76e933fbba4adf4e3a4a815bd7c8e","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a509b0a8ccb76e933fbba4adf4e3a4a815bd7c8e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-09 12:52:01.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aaaaff35ab3f7faeedcc3180f0f167c8fb53e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aaaaff35ab3f7faeedcc3180f0f167c8fb53e5a"}]},"branch":"refs/heads/master"},"111a6251958dd7be8a13ad3a8a9e2a1b711aa09d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-08-09 15:39:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/10"}}},"commit":{"parents":[{"commit":"960b4f59438554fb29cb7cf4661eaf895525236e","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/960b4f59438554fb29cb7cf4661eaf895525236e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-09 15:30:41.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111a6251958dd7be8a13ad3a8a9e2a1b711aa09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111a6251958dd7be8a13ad3a8a9e2a1b711aa09d"}]},"branch":"refs/heads/master"},"60b07d7e5397d29918abe12e948a8b0de3368dcc":{"kind":"REWORK","_number":11,"created":"2021-08-10 13:24:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/11"}}},"commit":{"parents":[{"commit":"960b4f59438554fb29cb7cf4661eaf895525236e","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/960b4f59438554fb29cb7cf4661eaf895525236e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-10 13:20:54.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60b07d7e5397d29918abe12e948a8b0de3368dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60b07d7e5397d29918abe12e948a8b0de3368dcc"}]},"branch":"refs/heads/master"},"8c46c09ad5d4ae6b87400b678a2bd7afc4af6631":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-08-10 14:27:34.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/788363/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/788363/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/788363/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/788363/12"}}},"commit":{"parents":[{"commit":"b10f71e42957f55f613130c545b37ab740109bba","subject":"mypy: create_volume flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b10f71e42957f55f613130c545b37ab740109bba"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:07:31.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-10 14:26:39.000000000","tz":-240},"subject":"mypy: image cache","message":"mypy: image cache\n\nChange-Id: Iebb002cba51fc0f5e42565151a1687da20bd5a24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c46c09ad5d4ae6b87400b678a2bd7afc4af6631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c46c09ad5d4ae6b87400b678a2bd7afc4af6631"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
