)]}'
{"id":"openstack%2Fcinder~788364","triplet_id":"openstack%2Fcinder~master~Iad0b5c7e689b2e524071322bfa59e789e1cc42a1","project":"openstack/cinder","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2021-10-27 21:16:28.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2021-10-27 21:16:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad0b5c7e689b2e524071322bfa59e789e1cc42a1","subject":"mypy: coordination.py","status":"MERGED","created":"2021-04-27 21:46:11.000000000","updated":"2021-10-27 21:18:49.000000000","submitted":"2021-10-27 21:16:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788364-mypy","meta_rev_id":"87197531dae9676f0ea6d125411fc5634632d210","_number":788364,"virtual_id_number":788364,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-27 21:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-10-27 17:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-10-27 21:18:49.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-27 21:18:49.000000000","post_submit":true,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-10-27 17:47:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 22:08:26.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 22:08:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-27 22:09:51.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-04-27 22:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 23:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 00:19:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 01:25:11.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 02:42:51.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-04-28 02:57:58.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 04:33:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-28 06:58:57.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 16:41:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 14:03:40.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-01 05:40:58.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-10 21:06:40.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-10 22:29:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-10 22:41:59.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-10 23:42:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-05-11 11:18:47.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-05-11 13:03:01.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-05-11 18:59:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-05-11 20:52:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 15:56:42.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-05-30 04:43:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 16:21:10.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 18:48:24.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-07-16 03:47:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 07:05:37.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-07-16 07:11:13.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"CC"},{"updated":"2021-07-16 08:27:58.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-07-28 16:01:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-08-04 16:07:35.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-08-11 07:45:04.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-10-27 17:47:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"5323498d044af41bddedc4cce0772f931020b5f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-27 21:46:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4539c29804f5b2d2b87b5a27eda954ec377f073e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-27 22:08:26.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"2542420cd7628de73afa7e72141fbbcf934f3e3e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-27 22:08:27.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"14ffab8d03a5b86304594cf0e166298306e038b3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-04-27 22:09:51.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"bb455b6aa56604586dc2f514baeef0fd0b006f18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 22:16:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9801948f13274570adaa6f538907d9eb : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3c7e33e02da4bf5bc0f669e5726c107 : SUCCESS in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4d12d4322b1320c6581c6df803f5fcaad5efbe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 23:39:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71e96c57f735414ab607c7307be8361a : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d7f16cb47740a5931b21a695f6ccf6 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/910b896f88204162b8bee0d4485aad3f : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42889312f0894f808a34ec66102ffac0 : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8252c9f5ae4a2698ca134e89260ac7 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3670f973b6b74129bd82477ce29fa3bd : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/1c4174f7b00045cc8dc1a0b8943315b8 : SUCCESS in 1h 05m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cdbd5ebb256417488278b9cac812608 : FAILURE in 54m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53fd6b877a1844df8213cde613e445fe : SUCCESS in 12m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da34760ca42f450b8fec17565c36a641 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a30c52fe86ff41629907a8c527aadec9 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fdde664ffbd34e4087966fa30459c877 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/93664a255ef94e08af6c78324532092b : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2760fec306a6444eb26e29e5e313361b : SUCCESS in 7m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/27d08d890f0a4fa08d4664f980e0cead : SUCCESS in 1h 14m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/012703e9c44b4683a6def1bacb9e654e : POST_FAILURE in 21m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c2eb563273843a298402b8cc3444183 : SUCCESS in 1h 23m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6df24afdf1c476e9511b778570ba827 : SUCCESS in 1h 17m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/35876392eeb54c97afcf10a6e3064d43 : SUCCESS in 1h 12m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ee5fb1529e447b79a4b255a7e58835a : SUCCESS in 1h 11m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/793fdbff4517490ab6baee08a648e5a0 : SUCCESS in 1h 48m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e953a0f4c66740899bc3427b50411bc5 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f1b7aa356db7331b3e4aefd2c55e31517f9404a8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-28 00:19:43.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/1/check/tempest-dsvm-full-focal-py3/8e73ce7/ : SUCCESS in 1h 53m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/1/check/devstack-plugin-ceph-dsvm-focal-py3/0bcbad4/ : SUCCESS in 1h 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7255cff7311db42400c82338a0c2dd9fd1ae333c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-28 01:25:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/64/788364/1/check/cinder-storpool-tempest/d58be58/ : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a4fb6a4d1bf6f80b053e79be714838adfd3a321f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-28 02:42:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/788364/1/check/dsvm-tempest-cisco-zonemanager-job/8039cf4/ : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":1},{"id":"431921b57659ac311dffd2e870cbfbc45363f0bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-28 02:57:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/64/788364/1/check/iscsi-driver-nimble/6e5292d/6e5292d.html : FAILURE in 42m 53s","accounts_in_message":[],"_revision_number":1},{"id":"be5b7fd010b44fb25c680cbcfc86de004b7813ab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-28 04:33:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4dea9ca/ : SUCCESS in 2h 30m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/060e1a5/ : FAILURE in 2h 23m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fc2c04bf158af27148096718507589029a0ea34d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-28 06:58:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-1/ns5_nfs : SUCCESS in 4h 54m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-1/ns5_iscsi : SUCCESS in 4h 54m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7dfc7a3a7313f7b9e45b8595be467f4d2d435a0f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-28 16:41:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/1/20210428230818/NEC-FC : SUCCESS in 2h 13m 55s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/1/20210428230818/NEC-ISCSI : SUCCESS in 2h 32m 54s","accounts_in_message":[],"_revision_number":1},{"id":"96f51858c672214a0da385f2512ea70b8113ed1e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-28 19:01:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c67202da3405707eea9b198aab97b92c7f8b29d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-28 19:10:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c40f16e/ : FAILURE in 7m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d728f7e/ : FAILURE in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f04faeb7e0611ec7d850fd0575e9988b33728c00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:26:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3291884bd0d453080f1e713db44aa3d : SUCCESS in 17m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f43ca5ddda914cabae2ef0ea6597c0cb : SUCCESS in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d4d0de189b0955db6c7cd0cb79e7823af711543","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-28 19:33:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/64/788364/1/check/cinder-storpool-tempest/5948dcb/ : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c1b4b91dfe92f174f091ed940d8c1553d3bb2289","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-28 20:26:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/788364/1/check/dsvm-tempest-cisco-zonemanager-job/0806c9d/ : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f0bd12caf52276347adea855faa603982eedac15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 20:54:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/881dc6fa7c0747a29bb56a2783c52f4b : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1afd77b35e6475498b9565b8e1beb4b : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de1e1349919e49b5869ede8ddbf1271c : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b7816d5a7244cb89beffee2f7a50b7 : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/644cd0af5ccc4a1fae85ab5bf1861f61 : SUCCESS in 15m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d60c52c1a4a466191bc1a89dc4d5934 : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c0c395accf5e42829282c1299ad0d85f : SUCCESS in 50m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2524030c3324fa4b1a1a37f4aae5095 : FAILURE in 1h 38m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d0512ef3c380480698cb8ad3104f49f4 : SUCCESS in 14m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7fe94883f1c344ff9bcb86c9318b9959 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/208de48a33794e2d87a939ff006ff3a4 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/238cd0b412b24c708b447ebc487e1493 : SUCCESS in 8m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d3e6f73121b14d08a46f3be1f9b99e01 : SUCCESS in 36m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51e7cb762b5647698523029057ad9e95 : SUCCESS in 8m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83971a8c77144fc49284f6a1e381a9c9 : SUCCESS in 1h 12m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/751ec6732cba40a5abc858f483a90aad : POST_FAILURE in 25m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/727efc0515964b6cb30d7f459ad11fb0 : SUCCESS in 1h 11m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5bf40cf53f645bfb6ea49760ce3ac26 : SUCCESS in 1h 26m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/289758d105ea4f92ac7468cb98e99569 : SUCCESS in 1h 15m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a50fb4e17ce49f6850bd636e571f00a : SUCCESS in 1h 13m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ef7f5808837a4a0e9c65e891a7f32ff5 : SUCCESS in 1h 50m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cde144d2b03d4eb0b00bfe2ebb1cd7ab : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b977c69fd96e8589961a2809970f89e7f5d3a04b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-28 21:55:25.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ce808a11e43a496b93f953f8a96d7838 : DISK_FULL in 2h 34m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/1761d08bc0d34a71b04e1f12821acf18 : POST_FAILURE in 1h 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"012c15df2c37f4797190384cf09316eaf6b6205c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-30 12:26:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"405b0ace9be68f08bf2d28048b97dcb874f2352a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 12:53:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1deaf9aa59254d849e1c62788d80ffaa : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a18a16dc90fb4b948f69724b3964a890 : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c62da9432fcf07385a002bc7ec0c2c37a10620cc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-30 14:03:40.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788364/2/iscsi/ : FAILURE in 1h 32m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/788364/2/smb/ : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":2},{"id":"30dfa040f90957bc8ad202fae3fdb39d2f31fc27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:23:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/361fe7ccd8ae439e98f6bc305804173a : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/942ab3500c014157897436f6dbe47274 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3aebec33d0b49618e259fd084bd472e : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119fe2edbb7a4f17b435f7bcc012e6f9 : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b55e8ce526c1485793e8054a026fcf13 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59cf9347806f4ed6902fb87b62845695 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ccfee7c95e6744febd21f94967ad132b : SUCCESS in 1h 08m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e277e50d57cb4f8bb118821d42b566d5 : SUCCESS in 1h 12m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/710ddc56eee74b96ac1f4dae2eadfe37 : SUCCESS in 14m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45d3e90e08ea4dafac06f61320114131 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/441eb00d700a4470b619f86cfc1ac3b2 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0dabd9d161b4d92905743e5f1fe6530 : SUCCESS in 6m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7c6c9cf7ea0648eebd6250ad85f914e4 : SUCCESS in 37m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b954913763349979b03a7efd1682fc7 : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/873c3b5a963647bab1925a280439464e : SUCCESS in 1h 25m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/439ee8bb95d2433f8494dd2d85750f0a : FAILURE in 1h 26m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c29686c2556e41299c4b94619f0af5a2 : FAILURE in 1h 41m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9bd4ca811cfb4f758bfbb5e4a782c437 : SUCCESS in 1h 33m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93c566b002cb42c0937c903593493cc6 : SUCCESS in 1h 24m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/35fa4ba8188b41a399df70d782e8acf1 : SUCCESS in 1h 25m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e94553951b494d51b98a03ca28d0b596 : SUCCESS in 1h 51m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/464a903fea2f47ba87a945dfe4f5a3d4 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4920ea18b5ac8b3b2d7e5844ba4957029d3cf1f4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-30 15:05:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/788364/2/check/dsvm-tempest-cisco-zonemanager-job/fa9c4a7/ : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e64e0b84ea668df04bc24bb9624e77b90d82f8f8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 15:09:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/788364/2/check/iscsi-driver-primera/63b483c/63b483c.html : SUCCESS in 1h 33m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4c8984002ac31bdba0e227e0922a22c6cbfb47dd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-30 15:25:38.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/64/788364/2/check/DellEMC_Unity_FC/6524f55/DellEMC_Unity_FC/81 : FAILURE in 2s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/64/788364/2/check/DellEMC_Unity_iSCSI/2e72e92/DellEMC_Unity_iSCSI/77 : FAILURE in 7m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1f517759d8ffabbe61909f6e5b645c83c30228f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-30 16:04:23.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2cc7ec742f524d68a7c13cc5f99c779a : POST_FAILURE in 1h 47m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ef3eaeb977a14166810335544b3b1c94 : POST_FAILURE in 1h 33m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6e12682125415978bfc3c7943bd977fd0805ad45","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 16:14:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/788364/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0fb243b/0fb243b.html : FAILURE in 2h 44m 33s","accounts_in_message":[],"_revision_number":2},{"id":"206be4329ec57245c514b60b519037ce20d9aeb0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-30 16:55:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/dbed972/ : SUCCESS in 2h 11m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f9fe04e/ : SUCCESS in 1h 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"9e50a79de3791dc24c106cb94f0ee0255e38f2d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-04-30 17:53:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/64/788364/2/check/cinder-storpool-tempest/1bbbc96/ : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1777ad82fe40865d3d6863d22107095e32d369df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-30 18:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/64/788364/2/check/iscsi-driver-nimble/12fcae6/12fcae6.html : SUCCESS in 1h 50m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b07e53c03b68322ccc514393b45f394b35284ad0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-30 19:10:29.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/64/788364/2/check/DellEMC_XtremIO_FC/8a864eb/DellEMC_XtremIO_FC/96 : SUCCESS in 2h 07m 16s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/64/788364/2/check/DellEMC_XtremIO_iSCSI/c169284/DellEMC_XtremIO_iSCSI/96 : SUCCESS in 1h 54m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"578050d7f41d1f8514ef60ae6add079387c21753","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-30 19:45:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-2/ns5_nfs : SUCCESS in 5h 2m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-2/ns5_iscsi : SUCCESS in 4h 54m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"2c2ca6b9a83c1f875cde719f1c2b2922a3b58cf3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-01 02:40:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/2/20210501091634/NEC-ISCSI : SUCCESS in 2h 12m 04s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/2/20210501091634/NEC-FC : SUCCESS in 2h 23m 29s","accounts_in_message":[],"_revision_number":2},{"id":"93e72be9b9d01586ad04d1366406ff1e873c582c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-01 05:40:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/64/788364/2/check/DellEMC_PowerMAX_FC/4f03a8e/DellEMC_PowerMAX_FC/41 : SUCCESS in 5h 24m 41s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/64/788364/2/check/DellEMC_PowerMAX_iSCSI/f4cde0f/DellEMC_PowerMAX_iSCSI/35 : FAILURE in 6h 30m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"9a90747759a44fe9298ea7d48825b99d4568264c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-10 18:41:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"aa7abe4a6c88320ade4667e0909af697e0867a91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:08:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d46d227fefd14fad907f74f389d2375c : SUCCESS in 21m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8603cd03cf2417a98117b9f853a423c : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52ca36e97b9733f5c9bc5ca8d4b92bb5a118153d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-10 20:16:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788364/3/iscsi/ : SUCCESS in 56m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/788364/3/smb/ : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9176fe5847b74ccc6e5561f8fd67ac3c06c67ff0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-10 20:48:53.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/3/check/tempest-dsvm-full-focal-py3/8275544/ : SUCCESS in 1h 25m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/3/check/devstack-plugin-ceph-dsvm-focal-py3/399c88b/ : SUCCESS in 1h 32m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8bc15e191c888efcfa2aaad2a530fab763895a3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 20:50:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccd5401be80e4646bcb480cc02a4a2b1 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a84a17d4aa724ac593dea7c32bd447e5 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1215bc75b44072b2a0eaed08f4bebf : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c3fcf9329794180889a01b78bae5b0a : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6a5986e25154e7982487352816cb7db : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f695d71db7f42d6b2fe325898990992 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b8afcd9667b642c29a70a611d3a48f8d : SUCCESS in 1h 11m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f69a37c184243e1b9f9d06804c01faa : SUCCESS in 1h 10m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9a3a9e812574f5d862310001199009e : SUCCESS in 14m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87879f7964814c6db67cce8fd9b57648 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb72b16e92824f1aad8d9e90318adab9 : SUCCESS in 7m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ae3900d65c54bb48df09c9978069f9c : SUCCESS in 8m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/18754e04955745c6b06ba66df9d271e0 : SUCCESS in 42m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ac57d12db6c44b4b468d23ce70e0b6c : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c6b915b6dec4c50aa0c965742763409 : SUCCESS in 1h 18m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1a7f56ff8d142f595f540c3e9540c00 : SUCCESS in 1h 24m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/144d1e09ac2d4b718f788a76c7ad6e70 : SUCCESS in 1h 06m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/45a611293a404f19b6058bbd34f12588 : SUCCESS in 1h 15m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77b0d15ab4674431915f4b7857ce5062 : SUCCESS in 1h 15m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/32f9dec242834d0a9a9fe5a7e4fb1f1d : SUCCESS in 1h 12m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3455fc7157ec4654bfc707040fe32466 : SUCCESS in 2h 01m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35f50d371d00419cb17eb1ab21e7ec14 : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":3},{"id":"0489872f861aeb08a7c0f4ab0983fb090bf49025","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-10 21:06:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/64/788364/3/upstream-solidfire-check/cinder-solidfire-iSCSI/8cd4f9b/ : FAILURE in 12m 31s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"89ad75bd4ef7b0294ab339118f167c0436201608","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-10 21:49:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/64/788364/3/check/cinder-storpool-tempest/f252496/ : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":3},{"id":"7e1c4c6a4b0072ca561747e210fa569d74535f89","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-10 22:29:10.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/64/788364/3/check/DellEMC_PowerFlex/78be3a4/DellEMC_PowerFlex/77 : SUCCESS in 1h 26m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"bfbf13558f3f6f538d3cc408c192ac5e171febf0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-10 22:41:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/64/788364/3/upstream-check/cinder-cDOT-iSCSI/b6aca50/ : FAILURE in 1h 30m 23s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/64/788364/3/upstream-check/cinder-cDOT-NFS/0333747/ : FAILURE in 1h 28m 56s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/64/788364/3/upstream-check/cinder-cDOT-FCP/078a375/ : SUCCESS in 1h 48m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5a18061cd4e1b013ee195abde02cf079d1e552fd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-10 22:51:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/788364/3/check/dsvm-tempest-cisco-zonemanager-job/ddc223a/ : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":3},{"id":"be7f342d899a9350a87a3a5341ab4c7abf56d100","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-10 22:52:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/788364/3/check/iscsi-driver-primera/a257b36/a257b36.html : FAILURE in 1h 10m 13s","accounts_in_message":[],"_revision_number":3},{"id":"0a087fa42b7a2cee0c57899d9aea4e65493e38fe","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-10 23:42:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/64/788364/3/check/DellEMC_PowerStore_iSCSI/1cd0e2d/DellEMC_PowerStore_iSCSI/69 : SUCCESS in 50m 43s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/64/788364/3/check/DellEMC_PowerStore_FC/e9e1f10/DellEMC_PowerStore_FC/70 : SUCCESS in 45m 43s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"a719669d8b05c0e360f8ae9f48b1a44b465314c6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-10 23:51:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/42e7fa0/ : POST_FAILURE in 1h 39m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/99ebf5f/ : TIMED_OUT in 3h 03m 31s","accounts_in_message":[],"_revision_number":3},{"id":"26fa52ca7f678ac76820808c47e2ad70c0eacff1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 06:16:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/788364/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c4837c0/c4837c0.html : SUCCESS in 2h 29m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9ee7170c84253f94df88a90357dfd0b533e71904","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-11 06:33:12.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_nfs : SUCCESS in 5h 29m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_iscsi : FAILURE in 5h 15m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"7331f1dabd2b5b559ea25e446675e00b22181389","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 07:26:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/64/788364/3/check/iscsi-driver-nimble/f4444ce/f4444ce.html : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":3},{"id":"aef929f237df61b3506238f5dc109b00465a72b0","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-11 11:18:47.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788364-3/Seagate-iSCSI : SUCCESS in 42:40.42\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788364-3/Seagate-FC : SUCCESS in 37:25.73","accounts_in_message":[],"_revision_number":3},{"id":"6621ad534d7e74d5f7beec379d2bc2e3d4f1c972","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-11 11:21:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/64/788364/3/check/DellEMC_Unity_FC/b0dc638/DellEMC_Unity_FC/151 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/64/788364/3/check/DellEMC_Unity_iSCSI/03d94f6/DellEMC_Unity_iSCSI/139 : SUCCESS in 1h 40m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2e8523ebf4787374e2cb20a268a78567c3b6ffc6","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-11 13:03:01.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788364-3/Lenovo-iSCSI : SUCCESS in 49:39.67\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788364-3/Lenovo-FC : SUCCESS in 44:09.76","accounts_in_message":[],"_revision_number":3},{"id":"a9c37e55ea41d6bd29f1f18288feb4a6ec674037","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-11 18:59:31.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788364-3/HPMSA-iSCSI : FAILURE in 1:26:15\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788364-3/HPMSA-FC : SUCCESS in 33:49.67","accounts_in_message":[],"_revision_number":3},{"id":"7b990693bb9dc74de16eac79dd54d755d711dc9f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-11 20:52:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/64/788364/3/check/DellEMC_VNX_iSCSI/ebe02ed/DellEMC_VNX_iSCSI/93 : FAILURE in 3h 52m 19s\n- DellEMC_VNX_FC http://publiclogs.emc.com/64/788364/3/check/DellEMC_VNX_FC/6f409bd/DellEMC_VNX_FC/97 : SUCCESS in 1h 32m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"2854fb8572f652ce599c1c38c20dde81928d3a56","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-12 05:01:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/3/20210512090632/NEC-ISCSI : SUCCESS in 2h 43m 10s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/3/20210512090632/NEC-FC : SUCCESS in 4h 55m 13s","accounts_in_message":[],"_revision_number":3},{"id":"71612a0f958b20b46f247ccabcb78ef21fde84d1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-14 00:21:27.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_nfs : FAILURE in 54m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_iscsi : FAILURE in 54m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"5156b4854578747dcd274f256f5bea6c9ef01460","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-15 02:21:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_nfs : SUCCESS in 5h 3m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_iscsi : SUCCESS in 5h 7m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"c330b6806f0fe1f8e1aeea6b7852b586b1bafb1a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-05-21 15:56:42.000000000","message":"Patch Set 3: Code-Review+2\n\nThis looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"2fa98e8a3317aa89a92c632d474ad226cdb9d3d6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-30 04:38:43.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_nfs : SUCCESS in 5h 10m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-3/ns5_iscsi : SUCCESS in 5h 15m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8bd1dd494d797b6726fc3caf17bce9f13f941763","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-30 04:43:48.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"903bea0ef8ee8c9a098bddbfe370033c33fc6197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-13 13:53:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"861b58d9f58e619a8c01029d1dfe5a97fc995dcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 14:16:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a763cf91b5ee4c4e80fedd818b2d829a : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/360cb87f8b6043bdbf243d8afadf44a9 : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b5c417488eca1d7a483dce2f8504fc083730cb6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-13 14:27:29.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"147b42d8a482b591405a3f7e840cb4c0da9e861a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 14:54:32.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"a288d3fb28d161f5c4b7d56e3b82c5b878cffe21","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-13 15:02:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788364/4/iscsi/ : SUCCESS in 1h 07m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/788364/4/smb/ : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":4},{"id":"3a737cc5f0a70b22b7318faf8d12b62bdb85bc45","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-13 15:20:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/788364/4/check/iscsi-driver-primera/dbb6cb6/dbb6cb6.html : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7bdbf36bd2a027ee6bfe81ac6a007af16966cc1a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-13 15:35:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cc40e31/ : SUCCESS in 1h 40m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f17c072/ : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":4},{"id":"4c85c6ea6c8307cccbc4319f7be97e01876ab0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 16:01:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08945328c40c449084f62343a4d19600 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984a879b6e3d418aafaeea8948f60a38 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8faa6c681cef44a58ee7c6fe8755b793 : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c548bf8317048779a2480179426e19c : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98eb2624ae1f44feac9ef913eed9d4ce : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2d56e86cbaf049348dd56088c623bc10 : SUCCESS in 1h 15m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53e1801274df4dcbb66723b3a22fd012 : SUCCESS in 1h 08m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dfb4dd6bd4184e8fb4cfca3fe943a649 : SUCCESS in 13m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/35ff93a2f40c4afda6a51044350bc376 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b07fff05f8948d7a960f102ec95ebb9 : SUCCESS in 7m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b33be9c9971349eb954bbb20441c0dca : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/222a55d09af54e3b8ecc4f87ecadd9b5 : FAILURE in 37m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2942d9f380d445c48a9cd90a7d25b28d : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5dbab3eab3b14b5cb7e93a5353b422d0 : SUCCESS in 1h 36m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a0b649f47a52457eb682c2b9823c00ee : SUCCESS in 1h 32m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eec7d7ea3e8f4487894e0c0d5ae7df46 : SUCCESS in 1h 38m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dbcc2c561eb24ad39241e817cc828980 : SUCCESS in 1h 22m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/245af8585ea344bcae9b1c89d4663afa : SUCCESS in 1h 13m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77fbfe46e10445c2a9e16e1f3cc5c72e : SUCCESS in 1h 19m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91240bfb241b4d609b99bccedaa91ec3 : SUCCESS in 2h 03m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb1a0540fcfb4a67bf2e7b3a7703b4f1 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f6dd61bd8d960642967979288169b5849ad5bfce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-13 16:15:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/788364/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cdc2647/cdc2647.html : SUCCESS in 2h 19m 04s","accounts_in_message":[],"_revision_number":4},{"id":"525b71c27b5c03bca382cc4827bc41a4fa0e8648","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-13 16:21:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/64/788364/4/ : SUCCESS in 2h 22m 54s","accounts_in_message":[],"_revision_number":4},{"id":"f58c2ca367628007801ca507f5f736f03cb2de36","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-13 17:10:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/64/788364/4/check/DellEMC_Unity_FC/1ca8f84/DellEMC_Unity_FC/426 : SUCCESS in 1h 14m 02s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/64/788364/4/check/DellEMC_Unity_iSCSI/b11666e/DellEMC_Unity_iSCSI/401 : SUCCESS in 1h 44m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"45995174045b2c3f8f066fe0d946e1334f25d568","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-13 17:16:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/64/788364/4/upstream-check/cinder-cDOT-iSCSI/6a07628/ : SUCCESS in 44m 39s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/64/788364/4/upstream-check/cinder-cDOT-NFS/a8520b2/ : SUCCESS in 45m 27s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/64/788364/4/upstream-check/cinder-cDOT-FCP/83ef3ac/ : SUCCESS in 1h 13m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"91188ccdfbfac98e75055a56b5ee7017786e107a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 17:34:03.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"8209800c8837c59b0741dded02b7af6cb5103cc6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-13 17:38:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/64/788364/4/check/DellEMC_XtremIO_FC/603deeb/DellEMC_XtremIO_FC/493 : FAILURE in 1h 13m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/64/788364/4/check/DellEMC_XtremIO_iSCSI/a3e694a/DellEMC_XtremIO_iSCSI/483 : SUCCESS in 1h 54m 06s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"dad0ffb851b1780cd4401f5a743a2db171cbbf72","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-13 17:55:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/64/788364/4/upstream-solidfire-check/cinder-solidfire-iSCSI/b06daba/ : FAILURE in 1h 51m 34s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"46cc4d15d5d1fa021eb8592f756a0ee010de6703","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 18:21:34.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"34be19fc4632dad124e48c99b1eb8743f097dd1a","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-13 18:48:24.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788364-4/PVME-iSCSI : FAILURE in 9:29.35\n* PVME-FC http://os-logs.tristero.net/dell-ci/788364-4/PVME-FC : SUCCESS in 47:06.10","accounts_in_message":[],"_revision_number":4},{"id":"ce7778325b291d2bac786debf0828e947d6389c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-13 19:13:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-4/ns5_nfs : SUCCESS in 4h 33m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-4/ns5_iscsi : SUCCESS in 4h 38m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"4b4fbcf11d3c9b3a5ba390176e205495c4aa1bf7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-13 19:27:50.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"66ab803de8b81548d78f790bf8ce4ad961c44d5a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-13 19:35:50.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788364-4/Seagate-iSCSI : SUCCESS in 42:02.75\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788364-4/Seagate-FC : SUCCESS in 37:13.88","accounts_in_message":[],"_revision_number":4},{"id":"f3b3e1b99647ab608850fbe194f185ca08417d68","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-13 20:00:21.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788364-4/HPMSA-iSCSI : FAILURE in 58:50.08\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788364-4/HPMSA-FC : SUCCESS in 32:59.13","accounts_in_message":[],"_revision_number":4},{"id":"bf0dee77af9aeb323e6b38705684b849590f74ea","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-13 23:59:46.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788364-4/Lenovo-iSCSI : FAILURE in 2:22:32\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788364-4/Lenovo-FC : SUCCESS in 48:07.58","accounts_in_message":[],"_revision_number":4},{"id":"5bce3451c66c255caba20034de1154e20e8e401a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-07-15 18:02:01.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"768d03673d07b603ed39f002b9903d84fe800611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-15 19:29:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"dbd6a3d58b11f62266954fdf5ff2af5bf16df53c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-15 19:49:20.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"8868e9471692a4c90ad313f4c6dc5b69b1b08223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:51:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fecef07a5334395adb7e7c0ac507f8b : SUCCESS in 16m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2d8c2a5bea94b37b3057d074e22a713 : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"673c3ff0eb47b3988c34c6670092febcf4f57e35","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 20:56:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/788364/5/check/iscsi-driver-primera/b74812c/b74812c.html : FAILURE in 1h 03m 10s","accounts_in_message":[],"_revision_number":5},{"id":"afd7f7016e768265a79e5763fbf370ce72bb571d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-15 21:18:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/64/788364/5/ : FAILURE in 51m 12s","accounts_in_message":[],"_revision_number":5},{"id":"efcda9a63aaf6902b5c6eaf870592e657c36f103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 21:54:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7ea5b8705e4d11b08b419062aadcc4 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681cd260d58a44d2814b9c279232cb5b : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55b3e5e9ce0b47498d1b4a5c35832406 : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c20a724878e49a89db4b56f5f20769f : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76c9cb5310c4f53ac03e459c84e77ae : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/8c228cfdc3764246a78cf3f7aba7d0ad : SUCCESS in 59m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18c148c71dd94bda8610a3dcbda94d5f : SUCCESS in 1h 10m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98ca739a990a4829834adfdda6eb38c8 : SUCCESS in 15m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/210745e71c31401c9792393eb549a30c : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ddc3a73139a4c69a758a945dab02d94 : SUCCESS in 10m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4db8fc45ff5c4a8ba266a994d39c1f6e : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a2f972323714ecebdc9b2324fc8312b : FAILURE in 35m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/126855776013450699f750fa65be8bc3 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/82cfda93ffec48378ab15fd422d4b4af : SUCCESS in 1h 15m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c9ed25e91eb346958a3d07839aa2300b : SUCCESS in 1h 24m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf7954fdf33c470f99f7831c50a4aae3 : SUCCESS in 1h 36m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02b81f1673c34f30acac44b626c24e85 : SUCCESS in 1h 10m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4322c2a5431a473aaccd1eb9812072e1 : SUCCESS in 1h 11m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f2ea634b5c5e493788f4f72328b0fd3b : SUCCESS in 1h 21m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b790ef6d602c46ff83b982e99c06c79d : SUCCESS in 2h 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f24c5e83461e45cba40aa024b194ca0d : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":5},{"id":"3cd008b985e80e6913989ec3297bd28b1b34628b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 22:33:34.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/4/20210716030111/NEC-ISCSI : SUCCESS in 2h 55m 38s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/4/20210716030111/NEC-FC : FAILURE in 4h 32m 19s","accounts_in_message":[],"_revision_number":4},{"id":"14e7df1cbacc887d683686bcec7bf1ccfed751f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-15 23:17:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/64/788364/5/upstream-check/cinder-cDOT-iSCSI/8cea367/ : SUCCESS in 37m 24s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/64/788364/5/upstream-check/cinder-cDOT-NFS/5252a2e/ : SUCCESS in 39m 08s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/64/788364/5/upstream-check/cinder-cDOT-FCP/3054d82/ : FAILURE in 1h 21m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"80f7ff772f1cee5a08bcbc43ee46f4241844afa5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-15 23:19:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/073ec51/ : TIMED_OUT in 3h 02m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9be0717/ : TIMED_OUT in 3h 03m 20s","accounts_in_message":[],"_revision_number":5},{"id":"ddc3b655a3122af84cb3e730e617fbdb236ca900","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-15 23:40:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/64/788364/5/upstream-solidfire-check/cinder-solidfire-iSCSI/c6e1d16/ : FAILURE in 1h 43m 11s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"78fe48dfc0dc7d5dbf685e489e9ea2f3834b885e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-16 00:11:07.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/64/788364/5/check/DellEMC_PowerStore_iSCSI/472a5a0/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/64/788364/5/check/DellEMC_PowerStore_FC/5c2970e/DellEMC_PowerStore_FC/249 : SUCCESS in 44m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"9ef99ffb6911c8c60fb6970a9b68babf2fb7392f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 00:34:03.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"a21a0626dc23f3accb912b8e465c5e4cf5cb2061","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-16 00:39:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/788364/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/76a61a5/76a61a5.html : SUCCESS in 2h 21m 09s","accounts_in_message":[],"_revision_number":5},{"id":"0d7f4469e536d24c8167f318a57327fb800ab027","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-16 01:39:50.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"c034d429acd57e7d09160c49a3b33964513c9f65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 01:43:22.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"b64e10b38d41ec67dbefba1281689fe4ba4d018a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-16 02:27:54.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/64/788364/5/check/DellEMC_Unity_FC/acc0b1b/DellEMC_Unity_FC/474 : FAILURE in 1h 18m 31s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/64/788364/5/check/DellEMC_Unity_iSCSI/5c025cf/DellEMC_Unity_iSCSI/445 : FAILURE in 1h 12m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c85cd07e35f9b46b6907eb72b13c0a199bc9faf0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-16 03:11:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788364/5/iscsi/ : SUCCESS in 59m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/788364/5/smb/ : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":5},{"id":"29401f0ccb222745329a61c0bb2c162ad3dae1cc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-16 03:39:31.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-5/ns5_nfs : SUCCESS in 4h 55m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-5/ns5_iscsi : SUCCESS in 5h 3m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"d9973ea1d7de202aed05ed0d14f2408cda4209f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 03:40:44.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"025ed14f7d7230404e687f3cab71191a18dc9603","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 03:46:23.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"00eded451cde937567bd640171b070ccb4679257","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-16 03:47:34.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788364/788364/5/check-not-vote/ext-cinder-zuul/5d738d5 : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":5},{"id":"96f9cee378403571edd5d760c146b683dcf72499","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-16 07:05:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3dded15a0b06a9169e123548a8e51a7db11bbd3b","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-07-16 07:11:13.000000000","message":"Patch Set 5:\n\nrun-DellEMC SC CI","accounts_in_message":[],"_revision_number":5},{"id":"46cbf9186d78907cd3f9039e2c4114ef18d224ab","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 07:15:03.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/64/788364/5/check/DellEMC_XtremIO_FC/fdec3ce/DellEMC_XtremIO_FC/538 : SUCCESS in 1h 18m 02s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/64/788364/5/check/DellEMC_XtremIO_iSCSI/06a3cb2/DellEMC_XtremIO_iSCSI/524 : FAILURE in 1h 54m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"c5e593608630f87e115598122dd28ffc0d806a5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 07:38:58.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"84416d8ca8ef64521c0a42df890b8acfed9a05a8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-16 08:27:58.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/64/788364/5/check/DellEMC_SC_FC/a0ed053/DellEMC_SC_FC/169 : SUCCESS in 46m 56s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/64/788364/5/check/DellEMC_SC_iSCSI/d4cc299/DellEMC_SC_iSCSI/165 : SUCCESS in 1h 16m 03s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"c7a67d2feeaa6dbea579e88bf0bc7a34c1965264","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-16 08:32:51.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788364-5/Seagate-iSCSI : SUCCESS in 43:55.80\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788364-5/Seagate-FC : SUCCESS in 39:48.36","accounts_in_message":[],"_revision_number":5},{"id":"c7d3f5110eb3b70188f111b9a6d1d84942213418","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-16 09:39:00.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788364-5/HPMSA-iSCSI : FAILURE in 59:26.77\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788364-5/HPMSA-FC : SUCCESS in 31:31.35","accounts_in_message":[],"_revision_number":5},{"id":"fdf67b4ce54b48356703ec5aa49b81492898d2c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 09:48:51.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"82e690e16b8b1231ea586cf414e77ad834193993","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-16 17:47:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/788364/5/check/dsvm-tempest-cisco-zonemanager-job/8ecf157/ : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":5},{"id":"6ee03d1ae05431b29ca4020bab7bb6ddfde57362","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-16 20:58:03.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788364-5/PVME-iSCSI : FAILURE in 3:01:06\n* PVME-FC http://os-logs.tristero.net/dell-ci/788364-5/PVME-FC : SUCCESS in 48:10.78","accounts_in_message":[],"_revision_number":5},{"id":"0554376c9991bcc09e1aeeafdde7639e7e65e29c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-17 02:28:26.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788364-5/Lenovo-iSCSI : SUCCESS in 50:09.75\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788364-5/Lenovo-FC : FAILURE in 48:57.89","accounts_in_message":[],"_revision_number":5},{"id":"8461a6f5f8180446e2b48b4aa278726208c4c247","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-18 06:19:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/64/788364/5/check/DellEMC_PowerMAX_FC/9fb7962/DellEMC_PowerMAX_FC/268 : FAILURE in 6h 49m 54s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/64/788364/5/check/DellEMC_PowerMAX_iSCSI/594a11f/DellEMC_PowerMAX_iSCSI/254 : FAILURE in 5h 52m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"b2f53d7a95041a6122952f9da4a3b8de2388d4c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-18 07:43:22.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"042a7254f519c34c73f65870ed41ab2cf473f720","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-28 16:01:11.000000000","message":"Patch Set 5:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/788364/5/fc : SUCCESS in 1h 03m 18s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/788364/5/iscsi : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":5},{"id":"dd31cc73e15847c32609a2a29243d3c80b907d9f","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-04 16:07:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLooks good +2","accounts_in_message":[],"_revision_number":5},{"id":"16e2242ce55b228808ac853d04936192764dd188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-09 13:14:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6ff18d7581e95fbfb3a4c5582732bd5b2911e7c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 13:52:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f74e81001c724940bd04383b617ad373 : SUCCESS in 17m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7903a93e53d0456db946bcefb23292c1 : SUCCESS in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a26d9ea12878333d8cd29a1ed4e18bad81e25393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:25:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85ac70f94534b2593fe7054977fcbaf : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de5928d0c5b412ea3da65e87f701188 : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa0aa650ef90415889725d5dc3dbfb22 : SUCCESS in 12m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a4a73e07b1c498ea933da2a3f9729ef : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121a55da05374b49b09bc65b47797426 : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/2f2880fed508474fbf185238e2fef8b3 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f163b045c83f4c75a2105b90dd9285de : SUCCESS in 1h 05m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea9cfb25216f499b9d2f04b7e37b81aa : SUCCESS in 12m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d0cadafb72794b3faa339337f73cf977 : SUCCESS in 6m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b54fec50c5674a9884b266190d9ece41 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9dcaafae6854a54a6f298a6f6b3edaa : SUCCESS in 7m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c42bfd7f39b344fa80e51d99c043cb8f : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/085f3520a76c48118b648c25b3b1b37d : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85017b4ab37c4150b74f34693416e7f3 : FAILURE in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eabab423ba22444b89652ed713eb3afc : SUCCESS in 1h 11m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b270c33ebf5a4f7ebfdf46382219473e : SUCCESS in 1h 26m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6598fcfdcd4a451da3dada545663dbfe : SUCCESS in 1h 26m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f21f8bf7c1194848a50313d1c08f46f5 : SUCCESS in 1h 15m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c5192f7a4294461b2679ca169b2ea9f : SUCCESS in 1h 18m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c79ceaecc77402f831fa578ccbc3aad : SUCCESS in 1h 10m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af90c76a1fbb4daabd2d8a1ad1d87ccd : SUCCESS in 2h 02m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19a84900e30145dfbd15f1c1a895c610 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":6},{"id":"b8b46884cb07b1ec53cc4666b618ac68e9371096","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-09 16:19:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/64/788364/6/check/cinder-storpool-tempest/6d2fb96/ : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":6},{"id":"0c5a16980c7ce3d16ae6519f6c45f67ccfdfab50","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-09 16:29:17.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/6/check/tempest-dsvm-full-focal-py3/6fdbf88/ : FAILURE in 2h 08m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/6/check/devstack-plugin-ceph-dsvm-focal-py3/59ce82a/ : FAILURE in 1h 10m 03s","accounts_in_message":[],"_revision_number":6},{"id":"68e8c3a362b3c7b976b5a1dde4d8461888599429","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-09 16:32:03.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788364/6/iscsi/ : FAILURE in 1h 49m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/788364/6/smb/ : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":6},{"id":"4df57c7da8d0f460a2edef39ff2cae56e9f6e9c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-09 16:49:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/788364/6/check/iscsi-driver-primera/5aea017/5aea017.html : FAILURE in 1h 03m 54s","accounts_in_message":[],"_revision_number":6},{"id":"3abf0e52d754869c1dcd390bd4c7a4107e96c28b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-09 16:50:25.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"53e08810431e546c8298adae0a04dfcc7fc37eaa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-09 16:50:25.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"2c30d50bf0cf0cc3d0d70d85d091955f5abd8fee","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:25.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"4f574900d6d3bd26fe3cb234bef0a1fccc6e5f02","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-09 16:50:26.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"3b03d98dfda804bb790192d38f3523c3a92e8c3c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:26.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"99feb6574ba0a1f383073f9b44fd66e868c26f19","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:27.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f53247e89b76e79a5140ce34b5a7ebad440730da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:29.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"7c04f3710abe82ab0aaccb814c70a5810f93fca6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-09 16:50:30.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"7335a1ef338725133b46fd7cd0aa0472e0fb0ecf","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-09 16:50:31.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"cdb7000f2306e779f97c4a0d00ddc7150ba4a914","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:31.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"02f1c7aee0f8201eb0180af8e30c0ef23bb8720a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-09 16:50:35.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"1b018721097267e60eb7f1eb14378d4693237faa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-09 16:50:46.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"62fef3615b6619f0766856fa0b3e11be81f74b21","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-09 16:57:36.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"833edb8d2281dcc86d25fc1f06419332d77cd2fb","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-09 23:34:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nLooks good to me","accounts_in_message":[],"_revision_number":6},{"id":"ae209806776f3669d3608dc38a5411c71d2a126b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-10 02:40:34.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-6/ns5_nfs : SUCCESS in 6h 40m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-6/ns5_iscsi : FAILURE in 6h 48m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"e76bbbac5a9b87c2fc83a9fbee41849093e6db4d","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-10 04:32:02.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/788364-6/PVME-iSCSI : FAILURE in 14:49.10\n* PVME-FC http://os-logs.tristero.net/dell-ci/788364-6/PVME-FC : SUCCESS in 47:06.25","accounts_in_message":[],"_revision_number":6},{"id":"ea9f234230fc44085b6c2dbd58b32d0eac0a20ff","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-10 08:35:06.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/788364-6/Seagate-iSCSI : SUCCESS in 41:50.23\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/788364-6/Seagate-FC : SUCCESS in 37:19.85","accounts_in_message":[],"_revision_number":6},{"id":"59c219a1eb4a093ff0dba0758314de79e3968fb6","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-10 11:16:10.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/788364-6/HPMSA-iSCSI : FAILURE in 59:14.77\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/788364-6/HPMSA-FC : SUCCESS in 31:32.30","accounts_in_message":[],"_revision_number":6},{"id":"f97b6cc95c2998cd33dbd5a22e0c6dc31fe4a4b4","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-10 19:32:35.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/788364-6/Lenovo-iSCSI : SUCCESS in 55:30.62\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/788364-6/Lenovo-FC : SUCCESS in 56:28.39","accounts_in_message":[],"_revision_number":6},{"id":"0394f7b20c08c3659e78e88a354f10e5cb9a8f79","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-08-11 07:45:04.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"50a9e97b0bcee2edba9c621138f2042088d56f5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-17 14:50:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"22dd6d422af87a7a4c88a06eb9906ef1322033a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 15:13:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4b8a1cab82243179b01f92b62b4db84 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c967911aeb2c4c83a9284cbe6eaf377a : SUCCESS in 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7f460a44a1d26e903ddbde77b71fe010eb238533","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-17 15:34:49.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/64/788364/7/check/cinder-storpool-tempest/2d26966/ : POST_FAILURE in 22m 19s","accounts_in_message":[],"_revision_number":7},{"id":"3636ddff03576ea1b26329140c5e0e79423e7768","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-17 16:08:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/788364/7/check/dsvm-tempest-cisco-zonemanager-job/df05c30/ : FAILURE in 30m 13s","accounts_in_message":[],"_revision_number":7},{"id":"ac069defc65c4a90ad0e297b35fa3db718b60b36","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-17 16:34:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/788364/7/check/iscsi-driver-primera/fc37ea4/fc37ea4.html : FAILURE in 1h 18m 09s","accounts_in_message":[],"_revision_number":7},{"id":"2290767a1d9d0b06c704f3bfdd922f237927bd9e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-17 16:37:07.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/64/788364/7/check/iscsi-driver-nimble/6021384/6021384.html : FAILURE in 37m 01s","accounts_in_message":[],"_revision_number":7},{"id":"929b096853ffca2e92d62b762d858d42281d276a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-17 16:50:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/788364/7/iscsi/ : SUCCESS in 1h 07m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/788364/7/smb/ : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":7},{"id":"cf5eeec0de5162ba706f03cff20bd3d0ac61a9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:55:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1b7c1a2733404fb9349ce0b18f9cab : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c08fd4bb5594e46b39f12c9cbbaf4d8 : SUCCESS in 12m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb3c68d71e042e29abdd9d2ebf2549f : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99a79915128c4e3a832c4b685ae2a9fd : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5390de2133a4e9a8c69473532f03981 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ff18cf27e3644cecb77d4c0954a33618 : SUCCESS in 1h 00m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eda249306d7843f9a320ff9c86be28e7 : SUCCESS in 1h 05m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b06556438fbf4aed9d63c7e9604715e9 : SUCCESS in 12m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b7e843be263940018bdb1ef33d0b57c8 : SUCCESS in 9m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6f2f6ebe88c4eed970179b241ed3c29 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/441e96d78ab644e5a227c6acf203e6b8 : SUCCESS in 8m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c9904137ef9474b9d841ef4af3c2467 : SUCCESS in 7m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea9724feaa7b481ab74b085a6a923c6b : SUCCESS in 40m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4a8b470edad0413e8b0cefa2e01771e0 : FAILURE in 6m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/badcf850b4974013b8f7b1546ccfdc61 : SUCCESS in 1h 11m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cda101fdd5744400ac045d3260d2dfd3 : SUCCESS in 1h 29m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78623e779ec14e68a539839f0d4b81b5 : SUCCESS in 1h 30m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fe5c0f8e7d1c44e8a4caa4e4255378d1 : SUCCESS in 1h 15m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e4a30b6ee6c144f183ea1c554bd69bf4 : SUCCESS in 1h 19m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34ebcecc43bb452caf7a1f882810d6a2 : SUCCESS in 1h 09m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a882d30209a4bf18201e513b21ce46a : SUCCESS in 2h 02m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fd7eb58ca634ffab7b08746aad8598a : SUCCESS in 47m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a724099fb14c4a53847b31190e130414 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":7},{"id":"6e02cc427502711be4c14d211c7e8ee6fe6b513f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-17 17:24:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/788364/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/81301c7/ : SUCCESS in 2h 27m 18s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/788364/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3516958/ : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":7},{"id":"7a05dc48cb5541d4bd06429cd7eecfeba7b8dcc5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-17 18:10:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/788364/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d65273d/d65273d.html : FAILURE in 1h 48m 56s","accounts_in_message":[],"_revision_number":7},{"id":"b3839fd8c3354cbb955c3bfa7e9774d01c94b370","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-17 18:18:00.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/788364/7/check/devstack-plugin-ceph-dsvm-focal-py3/e19f641/ : TIMED_OUT in 3h 10m 55s","accounts_in_message":[],"_revision_number":7},{"id":"81322df7e74f9943d8716ec94e13e1bc17aa9e60","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-17 18:42:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/64/788364/7/upstream-solidfire-check/cinder-solidfire-iSCSI/7e4d0dd/ : FAILURE in 1h 44m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"bb97eeec94d5cafe99926a043c903383a0f76694","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-17 19:07:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/64/788364/7/upstream-check/cinder-cDOT-iSCSI/c357993/ : SUCCESS in 36m 53s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/64/788364/7/upstream-check/cinder-cDOT-NFS/bc6d646/ : SUCCESS in 38m 50s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/64/788364/7/upstream-check/cinder-cDOT-FCP/e165656/ : FAILURE in 2h 11m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"3c21a472e1730f72714919c02b9e4dc815abe835","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-17 19:57:53.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/64/788364/7/check/DellEMC_PowerStore_iSCSI/8de1a80/DellEMC_PowerStore_iSCSI/355 : SUCCESS in 1h 30m 50s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/64/788364/7/check/DellEMC_PowerStore_FC/c778efb/DellEMC_PowerStore_FC/396 : SUCCESS in 1h 08m 15s\n- DellEMC_PowerStore_NFS http://publiclogs.emc.com/64/788364/7/check/DellEMC_PowerStore_NFS/e84e443/DellEMC_PowerStore_NFS/44 : FAILURE in 45m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"0bb576e7377cf0516476059d1e3ab3672d8ddb79","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-17 20:01:42.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-7/ns5_nfs : SUCCESS in 4h 9m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-788364-7/ns5_iscsi : SUCCESS in 4h 17m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"1eb7f7239164c49805840da42306226ae76838c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-17 20:01:58.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"a76e86858dba0ff4221832b62a50b9c554e78d8a","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-17 20:49:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/64/788364/7/check/dsvm-tempest-TOYOU-ACS5000-driver/5417a3e : SUCCESS in 1h 32m 07s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/64/788364/7/check/fc-test/75ac5bf : SUCCESS in 1h 41m 39s","accounts_in_message":[],"_revision_number":7},{"id":"10f0a25f9f5125b2e5e1073e907de7797233f951","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-17 20:50:08.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"1a5d15e90ea0120d9a85f2e059791feaafd4e162","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-17 21:28:36.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/64/788364/7/check/DellEMC_SC_FC/36d7be9/DellEMC_SC_FC/313 : SUCCESS in 50m 51s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/64/788364/7/check/DellEMC_SC_iSCSI/6f5d3d1/DellEMC_SC_iSCSI/303 : SUCCESS in 1h 53m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"fa1fbfd6a707d5b81609ca6c0959be43a4013b26","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-17 21:29:26.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"4107a0aaa6e23fe7036a5fc0aedee732a59caf78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-17 22:52:15.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"8fa9ae83be25054f558d1ef4c50d044dcd388a50","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-18 09:31:50.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/64/788364/7/check/DellEMC_PowerFlex/d5c041d/DellEMC_PowerFlex/405 : ABORTED in 3h 00m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"af3a492258a0bbb3f4df8197ec429ec6012084b1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-18 12:20:12.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/788364/788364/7/check-not-vote/ext-cinder-zuul/e6f789f : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":7},{"id":"d90a674b46d359c9d2f1472a2ed78debf3b31d9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-18 12:35:55.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/788364/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"6debfd9a79e15ae350147939e548ca6cb4adea00","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-19 07:19:12.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/64/788364/7/check/DellEMC_Unity_FC/77b5e18/DellEMC_Unity_FC/672 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/64/788364/7/check/DellEMC_Unity_iSCSI/7f07616/DellEMC_Unity_iSCSI/616 : FAILURE in 2h 58m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"a2c6ac3c041133e467fea5c88dab039f1244675b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-20 21:21:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/64/788364/7/check/DellEMC_XtremIO_FC/06d7897/DellEMC_XtremIO_FC/695 : SUCCESS in 1h 15m 54s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/64/788364/7/check/DellEMC_XtremIO_iSCSI/f914312/DellEMC_XtremIO_iSCSI/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"fdcb0901eb3760a4e5d62066ca1abee0edd8250e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-08-26 21:10:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/7/20210827013535/NEC-FC : SUCCESS in 2h 36m 41s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/788364/7/20210827013535/NEC-ISCSI : SUCCESS in 4h 35m 02s","accounts_in_message":[],"_revision_number":7},{"id":"5043bed0d2a575f1c538e1a74d66f9581c542b10","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-10-15 16:01:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d60616e9abf839a875393a11f135a7939296b601","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-10-27 17:47:40.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d2f090b1c1644fa15036284842405477a023aa87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 17:49:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"93a4e79b50caac561153ad87bc8b9ff64186ba05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 21:16:28.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04925f7d362644e791de30fa390c1a3c : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f172e1e026b441093d1827f98e848ea : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cc7fb65fb244835b89ad32c10e1b686 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6795d1217b14b91b37318c0b76c0c2b : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/cca218a0a9084539a9661c0e7367f920 : SUCCESS in 1h 09m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e24f4f05fd144646962a9fefa468a42d : SUCCESS in 1h 31m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8b91b77d4964d1daf42375aa1111d45 : SUCCESS in 1h 18m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b99d280aed1a430eb2884bd07eec1abe : SUCCESS in 1h 26m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2ca5fc206a84889a0cc6019d195bedd : SUCCESS in 1h 14m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11601d9b1ae348d883042e1adf4e1c86 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":7},{"id":"6648dab17e4b3f47d946577ae2bce097f42ead2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 21:16:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b295d9e0eefea255bb1ae8ac54cb477533098039","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 21:18:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f649cf2a0a7249b2a4c4fd547cb7fd7b : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"61335da5bc5e3cd158520fd5c3592e548ebcb32f","revisions":{"96e857726e254f3691a300eb24d83b172329ea0e":{"kind":"REWORK","_number":1,"created":"2021-04-27 21:46:11.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/1"}}},"commit":{"parents":[{"commit":"f2c6fa08497e99e3b96aa4f2c72b3c3be807cc41","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2c6fa08497e99e3b96aa4f2c72b3c3be807cc41"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 21:32:32.000000000","tz":-240},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96e857726e254f3691a300eb24d83b172329ea0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96e857726e254f3691a300eb24d83b172329ea0e"}]},"branch":"refs/heads/master"},"c843a730b5dd23d2a9541d1daf1cf00433d01b66":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-30 12:26:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/2"}}},"commit":{"parents":[{"commit":"d7593ca44679626407fbbfb278880913c03c5fbd","subject":"mypy: image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7593ca44679626407fbbfb278880913c03c5fbd"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-30 12:17:59.000000000","tz":-240},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c843a730b5dd23d2a9541d1daf1cf00433d01b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c843a730b5dd23d2a9541d1daf1cf00433d01b66"}]},"branch":"refs/heads/master"},"4808f78106b271080abd6ca11eac5d8ce0261b7b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-10 18:41:26.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/3"}}},"commit":{"parents":[{"commit":"db880b4a70cfe1945d4252afa055f7470710acce","subject":"mypy: continued manager, scheduler, rpcapi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db880b4a70cfe1945d4252afa055f7470710acce"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-10 18:36:05.000000000","tz":-240},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4808f78106b271080abd6ca11eac5d8ce0261b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4808f78106b271080abd6ca11eac5d8ce0261b7b"}]},"branch":"refs/heads/master"},"5c75cf06af501287e24f93e16e27811aefa65737":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-13 13:53:36.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/4"}}},"commit":{"parents":[{"commit":"fa529d09103f0c7367e480fdd2bae17312e1c4ad","subject":"mypy: continued manager, scheduler, rpcapi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa529d09103f0c7367e480fdd2bae17312e1c4ad"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 13:53:36.000000000","tz":0},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c75cf06af501287e24f93e16e27811aefa65737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c75cf06af501287e24f93e16e27811aefa65737"}]},"branch":"refs/heads/master","description":"Rebase"},"6181f25097a1498671b93f33fa6900a5464558ff":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-15 19:29:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/5"}}},"commit":{"parents":[{"commit":"ecf87d4eeebf834721405d2daf05b415e58285e9","subject":"mypy: continued manager, scheduler, rpcapi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecf87d4eeebf834721405d2daf05b415e58285e9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-15 19:29:20.000000000","tz":0},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6181f25097a1498671b93f33fa6900a5464558ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6181f25097a1498671b93f33fa6900a5464558ff"}]},"branch":"refs/heads/master","description":"Rebase"},"b81484dbc54e8f60bf6da9c5ebc56c572c0b7c3c":{"kind":"REWORK","_number":6,"created":"2021-08-09 13:14:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/6"}}},"commit":{"parents":[{"commit":"856242f8c90d7bdcb5a633cbfe2001f026167b7d","subject":"mypy: continued manager, scheduler, rpcapi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/856242f8c90d7bdcb5a633cbfe2001f026167b7d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-09 13:13:46.000000000","tz":-240},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b81484dbc54e8f60bf6da9c5ebc56c572c0b7c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b81484dbc54e8f60bf6da9c5ebc56c572c0b7c3c"}]},"branch":"refs/heads/master"},"61335da5bc5e3cd158520fd5c3592e548ebcb32f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-08-17 14:50:46.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/64/788364/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/788364/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/788364/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/788364/7"}}},"commit":{"parents":[{"commit":"b5ac2af0c2910b5d8c87a2b6dbd9b6f5f23a7bca","subject":"mypy: continued manager, scheduler, rpcapi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5ac2af0c2910b5d8c87a2b6dbd9b6f5f23a7bca"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-27 17:08:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-17 14:50:46.000000000","tz":0},"subject":"mypy: coordination.py","message":"mypy: coordination.py\n\nChange-Id: Iad0b5c7e689b2e524071322bfa59e789e1cc42a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61335da5bc5e3cd158520fd5c3592e548ebcb32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61335da5bc5e3cd158520fd5c3592e548ebcb32f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
