)]}'
{"id":"openstack%2Fcinder~789077","triplet_id":"openstack%2Fcinder~master~I994af226743a50f11d89373868e588d0d0313785","project":"openstack/cinder","branch":"master","topic":"drop-l-c","hashtags":[],"change_id":"I994af226743a50f11d89373868e588d0d0313785","subject":"Drop lower-constraints jobs","status":"MERGED","created":"2021-05-02 02:30:37.000000000","updated":"2021-07-12 17:19:25.000000000","submitted":"2021-07-12 17:16:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789077-drop-l-c","meta_rev_id":"3d397e3337664e2a616fc524665ab6d8f9673007","_number":789077,"virtual_id_number":789077,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 17:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-07-12 15:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-12 15:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2021-07-12 15:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-07-12 15:32:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-12 15:40:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-07-12 15:22:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-02 02:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 03:45:42.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 03:46:59.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-02 03:53:16.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-02 03:59:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-02 04:00:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 04:04:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-02 04:37:53.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 04:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-02 04:41:58.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 04:42:23.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-02 04:54:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 05:06:31.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 05:38:54.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-05-02 05:38:55.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-05-02 06:08:49.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-05-02 06:50:47.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 07:07:38.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-05-02 07:10:26.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-05-02 08:31:03.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-05-02 09:18:22.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-02 14:59:48.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-03 00:56:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-05-03 15:13:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 02:08:36.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-07 20:22:46.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 15:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-07-12 15:23:57.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-07-12 15:40:18.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"}],"messages":[{"id":"5203a346bbcf8c0a13d98e81cc28d8919cafee25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-02 02:30:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"585822d4a56f9a545d118c7622aa9d7feb5ead15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 02:52:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f471b48c01c4501b31b95d9ef5f99a1 : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c77ff6b6e2d14970b723f1314022dca7 : SUCCESS in 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82d5490a4d19ed42488c69c8fecf19a1e712a886","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-02 03:45:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/77/789077/1/check/dsvm-tempest-zadara-driver/0ceb041 : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f015e8aeaa7932b382a7d1e3d8fd38262feb46e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-02 03:46:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/77/789077/1/check/dsvm-tempest-cisco-zonemanager-job/54a3566/ : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fb53850d2c4daffeee570f9f9f9a9f55f11fb20c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-02 03:53:16.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789077/1/iscsi/ : FAILURE in 1h 21m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/789077/1/smb/ : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e500dc8e2ff3743088d2075ce9625582eb57bb3b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-02 03:59:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/77/789077/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0ee0265/ : SUCCESS in 1h 26m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/77/789077/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8d22992/ : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"678642b7f8f4be7767d1cc057a84bfdb61221336","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-02 04:00:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/77/789077/1/check/iscsi-driver-primera/2dfcdbb/2dfcdbb.html : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3a19f80cc63988a47c84ec0a4a44dd1a79737d94","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-02 04:04:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_Unity_FC/c7602be/DellEMC_Unity_FC/94 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_Unity_iSCSI/2425979/DellEMC_Unity_iSCSI/88 : SUCCESS in 1h 32m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"809226928a8a37f115d45ee35d9c504df26b7c96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-02 04:19:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/77/789077/1/check/iscsi-driver-nimble/2bac9d9/2bac9d9.html : SUCCESS in 1h 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5fdb8b67edb6164f84cab170c1be9d2944071a27","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-02 04:37:53.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_XtremIO_FC/cecb0d0/DellEMC_XtremIO_FC/107 : SUCCESS in 2h 06m 46s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_XtremIO_iSCSI/2900e7d/DellEMC_XtremIO_iSCSI/107 : SUCCESS in 1h 54m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ec13f4f2488888f95a4105e2af56ba3d7935367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 04:40:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7239ad2b4c9d40ab890887ea8e7ecd24 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba0a97b94a44b33859a734bc99b63c0 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da82dc20dab44669419de6372321070 : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf2932d5de344b6ca2000cfd77ca40a7 : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6437a7e68074aa2a71d46b0b0b94b2e : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e5976c1397483c839f41bf9135228e : SUCCESS in 11m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4555ca1ce7fb48f2ae10390aa958a95e : SUCCESS in 4m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/34f36a1275164d0e902a31db230f4aae : SUCCESS in 1h 04m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33464614b1424514bf5e30e0fa5e26d9 : SUCCESS in 52m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfbf2ebc87a64e24bc3c400e2107b9f6 : SUCCESS in 8m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f9d4f8148a34096a463d9f3e5f37236 : SUCCESS in 12m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd6a09fde3bc4a9a88d29bd544a992aa : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cdb626b23532405081774dfcf83c0cfd : SUCCESS in 10m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab2fa79032ee4260bfe93d58bac46682 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11c56c0f79564251855761ba2b499d56 : SUCCESS in 33m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/523a2f246a8d4e6ba98f1ad9ef5791cb : SUCCESS in 4m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb8d4be84dd542d1becc1ece6740be15 : SUCCESS in 1h 03m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ca175510866412eaa4534856feeb422 : FAILURE in 1h 19m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f5d6a1cea994976b0a7692072eee63a : SUCCESS in 1h 28m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc27d453550748a5a6ba8323f36f18bd : SUCCESS in 1h 10m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f71701bfef35478ba8fbf26afc8035ed : SUCCESS in 1h 19m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/24ecf766543a4ab8b111409d43b11c8e : SUCCESS in 1h 06m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/796f7e40870d4f2c960e4e2ebda9459a : SUCCESS in 2h 07m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4aa6714159e464188b9a95e9d7b9984 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":1},{"id":"54fda551014474bf88351c22d7265eca4571a271","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-02 04:41:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/77/789077/1/upstream-check/cinder-cDOT-iSCSI/b74a712/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/77/789077/1/upstream-check/cinder-cDOT-NFS/096f32c/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/77/789077/1/upstream-check/cinder-cDOT-FCP/36e682b/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"675b5dbbed6de0fb71dacdd87cce95e7a28b43ec","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-02 04:42:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/77/789077/1/upstream-solidfire-check/cinder-solidfire-iSCSI/278434d/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6daedf6c8ca51477487953048c9115c507731a1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-02 04:50:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/77/789077/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d854a08/d854a08.html : SUCCESS in 2h 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d48154426babe65e9ff0bc8795c199bc6e4bdf3f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-02 04:54:18.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5ee769154f1f45fead3798b3a015c0c8 : POST_FAILURE in 1h 36m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/52adda5ecadd4b339c0a24c60b6e79f3 : POST_FAILURE in 1h 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"681b937d1740995e0deb6e5dc9151a6d5629360d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-02 05:06:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/789077/1/20210502113041/NEC-FC : SUCCESS in 2h 13m 26s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/789077/1/20210502113041/NEC-ISCSI : SUCCESS in 2h 35m 47s","accounts_in_message":[],"_revision_number":1},{"id":"247d5bc68bc99f6700fb4bcb1f50db2571b62961","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-02 05:38:54.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/789077-1/Seagate-iSCSI : FAILURE in 15:40.36\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/789077-1/Seagate-FC : FAILURE in 15:35.45","accounts_in_message":[],"_revision_number":1},{"id":"7d17d633456ac75a31a0a6488613d96d31e47d9c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-02 05:38:55.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/789077-1/Lenovo-iSCSI : FAILURE in 15:32.20\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/789077-1/Lenovo-FC : FAILURE in 15:31.02","accounts_in_message":[],"_revision_number":1},{"id":"eb3416fed8e272c9755d9b1113e25653fda9bbd8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-02 06:08:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_PowerStore_iSCSI/4ee5e04/DellEMC_PowerStore_iSCSI/39 : SUCCESS in 49m 15s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_PowerStore_FC/9fbb303/DellEMC_PowerStore_FC/40 : SUCCESS in 47m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"3602c6cd84615566fb6dbc08b78ced5fdfc93397","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-02 06:50:47.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-789077-1/ns5_nfs : FAILURE in 53m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-789077-1/ns5_iscsi : SUCCESS in 4h 19m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"858bc1af7908a97d116fd09b44b150377e978aad","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-02 07:07:38.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/789077-1/HPMSA-iSCSI : FAILURE in 1:26:26\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/789077-1/HPMSA-FC : SUCCESS in 33:36.62","accounts_in_message":[],"_revision_number":1},{"id":"7540790abf2c3721d510eeb219f084207558f7cb","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-05-02 07:10:26.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/789077-1/PVME-iSCSI : SUCCESS in 1:05:37\n* PVME-FC http://os-logs.tristero.net/dell-ci/789077-1/PVME-FC : SUCCESS in 56:06.91","accounts_in_message":[],"_revision_number":1},{"id":"ae641103d556e0f00752854450d83f548aba155f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-02 07:46:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_Unity_FC/e8d0a1c/DellEMC_Unity_FC/95 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_Unity_iSCSI/8250bc7/DellEMC_Unity_iSCSI/89 : SUCCESS in 1h 35m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"188b3488b8ea967f600314173de752d0af2df7a8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-02 08:17:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_XtremIO_FC/575bfc2/DellEMC_XtremIO_FC/108 : SUCCESS in 3h 37m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_XtremIO_iSCSI/3bf59bd/DellEMC_XtremIO_iSCSI/108 : SUCCESS in 1h 53m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"44eb316855b1f37c173ac7a792cfd46352fa77cf","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-02 08:31:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_SC_FC/61d5537/DellEMC_SC_FC/62 : SUCCESS in 1h 54m 48s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_SC_iSCSI/41568f1/DellEMC_SC_iSCSI/59 : SUCCESS in 1h 57m 47s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b5b80175359bc42bf82fc8872014a687e547cbb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-02 09:18:22.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_VNX_iSCSI/89865fc/DellEMC_VNX_iSCSI/66 : SUCCESS in 2h 59m 26s\n- DellEMC_VNX_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_VNX_FC/629411e/DellEMC_VNX_FC/69 : SUCCESS in 1h 18m 32s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"dd2989e26c8f51864d0b0944c541f25235c0e836","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-02 14:59:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/77/789077/1/check/DellEMC_PowerFlex/c6cbc98/DellEMC_PowerFlex/41 : SUCCESS in 2h 37m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2d1a76bec2fe0dea605cd511045eef7d33a57e4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-03 00:56:33.000000000","message":"Patch Set 1: Code-Review-2\n\nThis requires more discussion.  We aren\u0027t planning to do this in master.","accounts_in_message":[],"_revision_number":1},{"id":"e401246ca7f44f9a691e9d07717eba7b119cf264","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-03 01:13:17.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-2\n\u003e \n\u003e This requires more discussion.  We aren\u0027t planning to do this in master.\n\nOk, got it, thanks for review","accounts_in_message":[],"_revision_number":1},{"id":"0f19a3713e4d4fdd4f9035e881b10125b45910b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-03 15:13:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/77/789077/1/check/cinder-storpool-tempest/7633677/ : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e72c7c2b0866a11ec81853e29b18832781602dc3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-03 18:39:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-789077-1/ns5_nfs : SUCCESS in 4h 39m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-789077-1/ns5_iscsi : SUCCESS in 4h 29m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a350cc9791d2fe63c31e09ae9a58833e19e5ea4e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-04 02:08:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/77/789077/1/check/DellEMC_PowerMAX_FC/00e3db2/DellEMC_PowerMAX_FC/52 : SUCCESS in 5h 13m 47s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/77/789077/1/check/DellEMC_PowerMAX_iSCSI/067ef4a/DellEMC_PowerMAX_iSCSI/46 : FAILURE in 5h 51m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e6a4b0f64682e4fafeafb37135907ddc0a55e5f5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-07 20:21:27.000000000","message":"Patch Set 1: -Code-Review\n\nWe discussed this at today\u0027s cinder meeting and decided to proceed with it:\nhttps://meetings.opendev.org/meetings/cinder/2021/cinder.2021-07-07-14.00.log.html#l-64","accounts_in_message":[],"_revision_number":1},{"id":"1b9e816bab39cc103994689a7678c407d6c6ec85","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-07 20:22:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3180587dc16d13bd0435b72ffb71f1dca94a0dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:22:19.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"d9dacd977628108e44bc6bdf98beb5cc40108ad4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:22:21.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"588b7375d0920c255a1239666f755806651d62fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-07-07 20:22:43.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5a9bb293000a63b2175b5f9f4c688a24f7083372","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-07 20:22:43.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b46ee18544ba48c2e03a56e94ced6bc83e524cd6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-07 20:22:44.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"dba8ef2dc388cec71931e133e7eb6b666d75ecd0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-07 20:22:46.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"8773be6de22a1a146d485269df8984aa6b77514c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-07 20:30:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e47095fb14c09df02fb6f9008760156b41cbd7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:49:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41a40970afbc48779ffb75e93f572604 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00c623b8de384a31b50ae7a44533c692 : SUCCESS in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b71512e8133f316dd3d60df7d941cd1f7585697e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-07 21:36:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/77/789077/2/check/iscsi-driver-primera/de12de1/de12de1.html : FAILURE in 1h 04m 32s","accounts_in_message":[],"_revision_number":2},{"id":"4beb9b351fa782789ca366f27460d5bfcbedd575","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-07 21:43:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789077/2/iscsi/ : SUCCESS in 1h 11m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/789077/2/smb/ : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a974ac5d1d95112dd9311383812f0a3946b57830","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-07 22:04:27.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/199d79a4c50a44cfa02fe7bb883f2b26 : POST_FAILURE in 1h 29m 49s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"0123e12a2987898aa8c572aeb5689c513a5b7aca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-07 22:07:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/77/789077/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/31ad42e/ : SUCCESS in 1h 24m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/77/789077/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3ce09bd/ : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3ce28e9da365c7bc3e61c0e3b11c17578ab8e52c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-07 22:41:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/77/789077/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5eeee10/5eeee10.html : SUCCESS in 2h 09m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f32910eec226a6be6761613fa1a9bea260c729e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:35:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5118c4ed515d45399c7f4fa2b7be7c0f : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58da539e4a3e4b0db55dd4609075fee5 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/196e582901ba47678944e87a4b989a8f : SUCCESS in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c903f383d09344daa8efc820d9a03c46 : SUCCESS in 13m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ae68cfa3d8b4201b52082d7e1b5b169 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95168bd27e8149b78d0f7b6c62b08233 : SUCCESS in 14m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/19381187d55445f5a3ae0ff13e897248 : SUCCESS in 4m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/5b128fea9d4b45bb877bfc10cb04225f : SUCCESS in 1h 00m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb9562071342445f8b361c4049ed2010 : SUCCESS in 1h 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63ebd8f754045cc83ef55c664a136b0 : SUCCESS in 7m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e94cf8fa61914833b2585f3ec33cf383 : SUCCESS in 11m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c446798455d442498edf24d08613b70 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5bd8352cfcd4833af5db9d6124bfd63 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/303a4ad2d46a4dae9e51023afe23d0ba : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ea8dec0e45446799861f7dab54cfc7a : FAILURE in 30m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81b1e3a3569e468aa5de98dd78a9b94b : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e474d99fce814565af60beda5b8ba635 : SUCCESS in 1h 11m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a3dc5e1e949749cdb38b197c3438730a : SUCCESS in 1h 31m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c5023ec34ab447a28d9270ceb8a32f58 : SUCCESS in 1h 37m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f33c82c2fc5647cb927b28f52c3a8763 : SUCCESS in 1h 13m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2c5b12a995c45d39c4b8b5b081334c4 : SUCCESS in 1h 10m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ae5638f160204177afae1725edf57760 : SUCCESS in 1h 19m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52b2cd6aacbd46979d0a6603518b5e74 : TIMED_OUT in 3h 03m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a4903cf5e5143509b4afc3107d20bf0 : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f5c095558df78b26cad0bef96952263090151f91","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-07 23:42:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/77/789077/2/check/dsvm-tempest-cisco-zonemanager-job/68f5969/ : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3a6c20595a8dbebf3f00928b945502e6dcea531e","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-07-08 02:03:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9bbe2703d0500c49a05cc4299a5fa0d4d9458803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 02:24:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c9b1ac26aa74ea083e7ebbc3e15a486 : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72c3a8f574f64ab59ecec07ddec8c99a : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"27ae9c0bb769a31bff20412c95121e2d329f56a6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-08 03:12:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/77/789077/2/check/dsvm-tempest-cisco-zonemanager-job/c7fc424/ : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":2},{"id":"baca00ab7d8448e9bd47a6c50816c9c54450e3f1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-08 03:26:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789077/2/iscsi/ : FAILURE in 1h 21m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/789077/2/smb/ : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c9097a0d0282b28928e5ae4960319dd1e2991bbf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-08 03:42:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/77/789077/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d7d95f2/ : SUCCESS in 1h 08m 58s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/77/789077/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bfb018b/ : SUCCESS in 1h 36m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b73663d3401a443c84f91c4f0817971ea5c07e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 04:02:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e0d1902396a64fafa1ae6b3a5a1b7223 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf95e433c33a4fbeb0434ccb98264a39 : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3910d7616fc4cec94b01c3ffeb68492 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e40fa2cf0eba447d8886ade37bfacce0 : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36b66c58c6f842e8a92be186bf52ce55 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085988bfda4c496086613fa83d618b3a : SUCCESS in 11m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eccb99dd1d374dd7ac907d1d0a5169a7 : SUCCESS in 3m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/ce93778238a446f4b4ceb7f468528b81 : SUCCESS in 55m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44dcdb6e30ac40f891ace259b025d1e9 : SUCCESS in 1h 15m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2fa1894d86142b380d18b4464ac3e8f : SUCCESS in 6m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5384ea3ae64444cdae16a9ff2fea2624 : SUCCESS in 12m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14ebba45122f45dca7d7b4dbc7f0f790 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea4acda719de444e98751f23308c6d2d : SUCCESS in 6m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/933f6e675bbe40aeb62e37b901022477 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6503e28232074b2a973c6ad45258ebe3 : FAILURE in 31m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9154022220d64e1786080f75cae49922 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2af10b9c8a647898cf90a09538e9719 : SUCCESS in 1h 21m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4ed935112f3e4b939df39bef8a229573 : SUCCESS in 1h 32m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/874c370b96f14529821b82fb7352dc2f : SUCCESS in 1h 27m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/484f5dffe12848ff965d7dc62bc0d4a7 : SUCCESS in 1h 15m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a87260bca0734c6c871975c2c81bc41c : SUCCESS in 1h 20m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5001713507a847ffa7db5dada894c6ba : SUCCESS in 1h 17m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d2b8eab3622b42ca91d8d1d8dc0214a2 : SUCCESS in 1h 56m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5907c28245c840cdb7d5894049f895c5 : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":2},{"id":"746039a5b68f849d65c629a1b94f533e5f5a5816","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-08 04:03:24.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/674d074765c94fab8255f0db6c174569 : POST_FAILURE in 1h 55m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"da9d73913514d353f7f49e76fb9bfb8a6302ce56","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-08 04:59:04.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/77/789077/2/check/DellEMC_PowerStore_iSCSI/e7880a7/DellEMC_PowerStore_iSCSI/215 : SUCCESS in 56m 33s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/77/789077/2/check/DellEMC_PowerStore_FC/3aff68f/DellEMC_PowerStore_FC/216 : SUCCESS in 50m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"2c29432cda5e71b1b48a3a052a844a083611ba4e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-08 05:01:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/77/789077/2/upstream-check/cinder-cDOT-iSCSI/0a2db88/ : SUCCESS in 37m 49s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/77/789077/2/upstream-check/cinder-cDOT-NFS/80f48cf/ : SUCCESS in 37m 09s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/77/789077/2/upstream-check/cinder-cDOT-FCP/1f91a62/ : SUCCESS in 58m 21s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0ad1e528ec1cd52c24c885f7f8bfa88c56bf8471","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 05:15:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/77/789077/2/check/DellEMC_VNX_iSCSI/960db93/DellEMC_VNX_iSCSI/245 : FAILURE in 1h 09m 53s\n- DellEMC_VNX_FC http://publiclogs.emc.com/77/789077/2/check/DellEMC_VNX_FC/df8b888/DellEMC_VNX_FC/263 : FAILURE in 53m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"007c9db20b06beb315f2e057f53550ad865af1b5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-08 05:48:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/77/789077/2/upstream-solidfire-check/cinder-solidfire-iSCSI/95113a0/ : FAILURE in 1h 44m 37s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"ab97a9aaabdab7a29196b3f1ce40cc8f0524858d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 06:17:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/77/789077/2/check/DellEMC_PowerFlex/740a5dd/DellEMC_PowerFlex/284 : SUCCESS in 1h 49m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"777d1d6bac14e28d445fcd850bedad8df6b2e68b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-08 06:38:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/77/789077/2/check/DellEMC_Unity_FC/a736b13/DellEMC_Unity_FC/392 : SUCCESS in 1h 21m 38s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/77/789077/2/check/DellEMC_Unity_iSCSI/859cff9/DellEMC_Unity_iSCSI/368 : SUCCESS in 1h 44m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"712d82cee9750ea5be582d6eefa3bbaf4c72274c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-08 06:39:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-789077-2/ns5_nfs : SUCCESS in 4h 29m 6s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-789077-2/ns5_iscsi : SUCCESS in 4h 32m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"50799d1aacc606668d7ab15320a68e8cdfd92abb","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-08 06:44:37.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/789077-2/Seagate-iSCSI : SUCCESS in 43:44.71\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/789077-2/Seagate-FC : SUCCESS in 37:21.17","accounts_in_message":[],"_revision_number":2},{"id":"fda4575fef98edfd9abeaa8a8adf7a239c0598e9","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-08 07:16:51.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/789077-2/PVME-iSCSI : FAILURE in 9:18.94\n* PVME-FC http://os-logs.tristero.net/dell-ci/789077-2/PVME-FC : SUCCESS in 46:48.34","accounts_in_message":[],"_revision_number":2},{"id":"74022cbd74b41dc5f5a77057559f141ec2d72334","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-08 08:38:19.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/789077-2/HPMSA-iSCSI : FAILURE in 1:02:19\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/789077-2/HPMSA-FC : SUCCESS in 33:37.81","accounts_in_message":[],"_revision_number":2},{"id":"eb036cdfea27d517efc10e7b48c67ad11f5422a6","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-08 08:48:30.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/789077-2/Lenovo-iSCSI : SUCCESS in 49:56.73\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/789077-2/Lenovo-FC : SUCCESS in 45:56.27","accounts_in_message":[],"_revision_number":2},{"id":"b289de5c0a3bdd46a497a6840064932a79aa0e78","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 11:19:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/77/789077/2/check/DellEMC_XtremIO_FC/4d4aebf/DellEMC_XtremIO_FC/460 : FAILURE in 1h 26m 38s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/77/789077/2/check/DellEMC_XtremIO_iSCSI/c545ac6/DellEMC_XtremIO_iSCSI/454 : SUCCESS in 1h 49m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"4caa1b10625e8560eb502ab55a258989347ad929","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-08 12:05:02.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/789077/789077/2/check-not-vote/ext-cinder-zuul/75e660e : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"74f843bcaf5b10251020ed40cac56ff407226b62","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-08 15:00:05.000000000","message":"Patch Set 2:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/789077/2/fc : SUCCESS in 0h 54m 06s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/789077/2/iscsi : FAILURE in 0h 24m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0970f28badb39a8b5e9ad20df072ec3895fb84be","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 20:33:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/789077/2/20210709025458/NEC-ISCSI : SUCCESS in 2h 35m 02s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/789077/2/20210709025458/NEC-FC : SUCCESS in 2h 38m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4b2e1fc8e5aa7d8d4da209481a7cfdf4a8c81aec","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-12 15:22:44.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nJust realized that this isn\u0027t really my patch, so +2 from me!","accounts_in_message":[],"_revision_number":2},{"id":"eeb1f5e06ffdc712b5a149fc3d490589fa9c0203","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-07-12 15:23:57.000000000","message":"Patch Set 2: Code-Review+2\n\nThe whole requirements/constraints thing is not my area of expertise, but I think this looks ok to me.","accounts_in_message":[],"_revision_number":2},{"id":"fbc9a37b589bcb7ef903360c9171064ca73d84e8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-12 15:32:25.000000000","message":"Patch Set 2: Workflow+1\n\nThis was discussed at the weekly cinder meeting with no objections, so no sense waiting to merge it.\nhttps://meetings.opendev.org/meetings/cinder/2021/cinder.2021-07-07-14.00.log.html#l-64","accounts_in_message":[],"_revision_number":2},{"id":"b4be148da5cd11cadad64b997327c308c3be78b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 15:32:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0689bfdc4f5ea370a490f08ac986d9f96f8fa1b3","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-07-12 15:40:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a8bab6d8f3281287faa1e2aeb80185aead49de23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 17:16:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a8f41be54202497d94cb048d0489b36c : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94db4c346fb745e3a25791401e304420 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3acc391e6557454f9554d6371fed08ef : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e693d13902d64ded809c5be7f1719ccf : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6700b3585d6b4edc851b7b774ff2903a : SUCCESS in 13m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7d4e5e312ba143fc8ea6984f9c2adcb3 : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/4b59369d3e4e474e88d6de83136e0f5f : SUCCESS in 1h 06m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/609e0423684644b4b56474406da62bb6 : SUCCESS in 1h 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe80e9701044ac0ac7844be62cd1bde : SUCCESS in 7m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f50482f30c8453aa07308fabff2f595 : SUCCESS in 1h 04m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/73d834c61ffd425487a7a86a20b86bf3 : SUCCESS in 1h 38m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea4f1264e008423694a61e0f163461fe : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d76546998ecf13f11d51a07fc639a14344cdbfd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 17:16:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3d397e3337664e2a616fc524665ab6d8f9673007","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 17:19:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3ec3f02bf4474df4b04a1bcd653f8391 : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7810405cb0354c47bb657193169abfd0 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd892b21866d42d5bfa6c5d53da215ba : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b604a3bf66829d87a7ebc7e165d24780cd0e4bd7","revisions":{"c3dc9b66d4fb62abd51950dd8fc2f507834a8b36":{"kind":"REWORK","_number":1,"created":"2021-05-02 02:30:37.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/77/789077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/789077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/789077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/789077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/789077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/789077/1"}}},"commit":{"parents":[{"commit":"2b7396e0749b3ea455d1d40bb5596f52b3855064","subject":"Merge \"zuul: add mypy experimental job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b7396e0749b3ea455d1d40bb5596f52b3855064"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-02 02:30:26.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-02 02:30:26.000000000","tz":480},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: I994af226743a50f11d89373868e588d0d0313785\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3dc9b66d4fb62abd51950dd8fc2f507834a8b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3dc9b66d4fb62abd51950dd8fc2f507834a8b36"}]},"branch":"refs/heads/master"},"b604a3bf66829d87a7ebc7e165d24780cd0e4bd7":{"kind":"REWORK","_number":2,"created":"2021-07-07 20:30:37.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/77/789077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/789077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/789077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/789077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/789077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/789077/2"}}},"commit":{"parents":[{"commit":"a7e98dba5b0bc84a2642952b66962d2e645c64f3","subject":"Merge \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e98dba5b0bc84a2642952b66962d2e645c64f3"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-02 02:30:26.000000000","tz":480},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-07 20:30:03.000000000","tz":-240},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: I994af226743a50f11d89373868e588d0d0313785\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b604a3bf66829d87a7ebc7e165d24780cd0e4bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b604a3bf66829d87a7ebc7e165d24780cd0e4bd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
