)]}'
{"id":"openstack%2Fcinder~789603","triplet_id":"openstack%2Fcinder~master~I2414e521eb0b4ce48b41dfe4d305e002e8de8f40","project":"openstack/cinder","branch":"master","topic":"bug/1926630","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-17 16:01:29.000000000","reason":"Their vote was deleted"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2414e521eb0b4ce48b41dfe4d305e002e8de8f40","subject":"Validate fields in encryption spec create/update","status":"NEW","created":"2021-05-04 16:21:36.000000000","updated":"2024-05-17 16:01:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":11,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"0cf17f42a091d66b4214ca7b088d595f124c5d04","_number":789603,"virtual_id_number":789603,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-15 17:22:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2021-07-15 20:38:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-08-10 06:25:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2024-05-17 16:01:29.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 16:43:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 17:23:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 17:36:36.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 18:36:35.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-04 18:52:32.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-04 19:26:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-04 19:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 20:41:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-04 20:49:59.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 21:53:10.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 21:55:02.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 01:17:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-08 06:38:26.000000000","updated_by":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"reviewer":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"state":"REVIEWER"},{"updated":"2021-05-11 20:00:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-11 20:00:18.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-05-11 22:06:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-11 22:55:53.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-11 23:08:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-05-12 00:51:08.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-05-12 00:56:09.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 04:46:31.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-05-12 05:06:51.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-05-12 14:42:52.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-05-12 16:56:39.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 20:06:21.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-06-09 02:02:27.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 23:08:23.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-10 03:09:09.000000000","updated_by":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"reviewer":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"state":"CC"},{"updated":"2021-06-16 21:51:53.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-06-19 20:23:47.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-06-21 05:17:03.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-07-07 20:07:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-07-09 12:00:30.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2021-07-16 18:44:10.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-19 02:23:05.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2021-08-10 06:25:15.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"}],"messages":[{"id":"b380485eb3ad5bb1e723341cdedba22e49a0cc40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-04 16:21:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd84939589d7f48fa68c64a360a0a967a1fa56aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 16:43:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3890feda959e4b14bfc76bd7ad36e82b : SUCCESS in 19m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94a46ed399364f4182d045c6e9032244 : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39f2e6f450a211a08b00942bf52cc1115bfb928d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-04 17:23:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/789603/1/check/cinder-storpool-tempest/52aaec2/ : FAILURE in 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e2c48760983390c8cbe2e9d92e8eb7e93f827359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-04 17:36:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"846e562c65c8e96509a5119163e33b81358cc440","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-04 17:36:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/1/check/iscsi-driver-primera/c0bbd06/c0bbd06.html : FAILURE in 1h 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3e9177ea3de6c318bc762527ffadb92c555de6b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 18:00:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28bc0a85a01e4ea69ceb13d2efd7782a : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c386f89fa9a4725a4caa0da872e14f1 : SUCCESS in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8eac009d180f09cbe6839afd2f61cea172277dc5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-04 18:36:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/2/iscsi/ : SUCCESS in 50m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/2/smb/ : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":2},{"id":"be2418bc31345845db36aaa9534a9677dc257d1d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-04 18:40:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/789603/2/check/cinder-storpool-tempest/32aab02/ : FAILURE in 29m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6b36eef40d0bd9bc5855940be3df654d5838c5de","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-04 18:52:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/2/check/dsvm-tempest-cisco-zonemanager-job/8debc0b/ : SUCCESS in 1h 10m 32s","accounts_in_message":[],"_revision_number":2},{"id":"624c70e890844882ee11b663e60be9ac9272cbfb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-04 18:54:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/2/check/iscsi-driver-primera/8b175d2/8b175d2.html : FAILURE in 1h 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"7b2e88335fcc435ee8e103e4c1b32e5e1a7b715c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-04 19:26:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/789603/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7c1d01d/ : POST_FAILURE in 1h 47m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/789603/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/959b544/ : FAILURE in 1h 48m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ad237fd4edaa2b8ee7cbdfa71c3c7945b88caaef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 19:27:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/451da24d88934e969ee838ed09a2bc34 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be8bc19f92b4ea89ab78e72e0af54d1 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd2f652b2a714e38b16d5f365edbe15e : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0797f78a74d40d8ae645d7be1445de4 : SUCCESS in 13m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fe378dff89a4b6c93ee9c23e3453e08 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32e20eea5684d3bbcd49687694768fa : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/d73a2c02e227405391735da4c57d28c4 : SUCCESS in 1h 11m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70746451458542438bf3b113c771316f : FAILURE in 1h 01m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fe01390fc42c4fcc931db844ec6445a9 : SUCCESS in 14m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14284daa4d7641b48254690e247ca5ec : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c29a372b735e47bc93438cd9c45f3847 : SUCCESS in 7m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6049e888725e40de8c38b671edf6335a : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/701d46d6d5ae4981935041e6427f3474 : SUCCESS in 37m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/677b2bfb46b84cd1a2686f8e1f16e5bd : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/295ac9105512421b9b79e89b1f38f432 : SUCCESS in 1h 26m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9af0719fe1fd44b698a0834d1c13eb5e : FAILURE in 1h 23m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20782a9eda414ab6b559133d44b81dfd : FAILURE in 1h 24m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d7ef13eac4942f5b5b94b9e28511cb2 : SUCCESS in 1h 21m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3050aa5c8f6648438d743fc076e80540 : FAILURE in 1h 23m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7227763d76764a69ae4ab6b1b7762da3 : FAILURE in 1h 06m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c896ab6dc494330b087e04abb3b5edb : SUCCESS in 1h 50m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7359a125734e4d66ae69888d1d5dbbde : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8805b83f3ff49037caf8fa140214fbe379c1fe40","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-04 20:34:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1d3dab0/1d3dab0.html : FAILURE in 2h 52m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8f4efd40a122ffc5e7e58ac4a516ee0fea6194ac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-04 20:41:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/2/check/DellEMC_Unity_FC/be97dab/DellEMC_Unity_FC/111 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/2/check/DellEMC_Unity_iSCSI/0fe4c7b/DellEMC_Unity_iSCSI/103 : SUCCESS in 1h 32m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"da2a064788ce8e65fb8c066d2da4223e1af0ec58","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-04 20:49:59.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/2/check/tempest-dsvm-full-focal-py3/e075af2/ : FAILURE in 1h 30m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/2/check/devstack-plugin-ceph-dsvm-focal-py3/91103f8/ : FAILURE in 1h 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a5a7dc77ace9e30ce85b08e7efae65b67cb9ff06","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-04 21:47:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/789603/2/check/iscsi-driver-nimble/6372f93/6372f93.html : SUCCESS in 2h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"17f066c0e032372610e6efba866be3ee1c04fb2b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-04 21:53:10.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-1/ns5_nfs : FAILURE in 4h 44m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-1/ns5_iscsi : FAILURE in 4h 45m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"708659462892902b32ba1843258392287267e193","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-04 21:55:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/2/check/DellEMC_XtremIO_FC/c6347d2/DellEMC_XtremIO_FC/120 : FAILURE in 2h 09m 54s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/2/check/DellEMC_XtremIO_iSCSI/aa9e8ec/DellEMC_XtremIO_iSCSI/120 : SUCCESS in 2h 00m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"a34801dcf7a3256ed6c572fa19368b023fa59b93","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-04 23:16:46.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-2/ns5_nfs : FAILURE in 4h 39m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-2/ns5_iscsi : FAILURE in 4h 35m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"facfa6ac32f70cc83c68d48d571615cdf44eafad","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-05 01:17:28.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/2/20210505054609/NEC-ISCSI : FAILURE in 4h 30m 29s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/2/20210505054609/NEC-FC : FAILURE in 4h 31m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b65fdd5df42a3eb225fcec8c239b3858ee0fbf40","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-07 14:50:28.000000000","message":"Patch Set 2:\n\ntempest\u0027s test_volume_type_encryption_create_get_update_delete currently creates invalid encryption specs and so this makes that test fail.","accounts_in_message":[],"_revision_number":2},{"id":"1599950e18a8c778b4d0db06eca89b8d5da6f413","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-08 06:38:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"695b68db88faa070adf5b3067718bd2eb94330e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-11 15:15:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"77aeff0b676099a87ec339476e164eeb4d0de44e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:37:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee5b274eff1c4c3eb3fd9fbc61e365f0 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce683b694c264886a9de58967727aba7 : SUCCESS in 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72bb692ad301906a1d834d61b676370bfb665911","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 15:42:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/789603/3/check/iscsi-driver-nimble/ae8606d/ae8606d.html : FAILURE in 25m 55s","accounts_in_message":[],"_revision_number":3},{"id":"39504c3680e3d81f65875f340ca220098b070af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-11 15:48:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"beddc1c078f08a8f524a349e32e0e3f23bc18d24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:14:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f882035444314f30b73d4755e76e4679 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfa8575d65eb43ffa34981effb7cf3d7 : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1f250eb8c765c46e6d8c6d0f222b5ce89fdf1565","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 17:12:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/789603/4/check/iscsi-driver-nimble/97e6c93/97e6c93.html : FAILURE in 25m 51s","accounts_in_message":[],"_revision_number":4},{"id":"b2f62bb2f0939c84f242d634b26500abf1bb4aca","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-11 17:12:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/4/iscsi/ : FAILURE in 1h 22m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/4/smb/ : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":4},{"id":"9c95f8b208773952748d71316e5698a28b971a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:50:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64885f4f95bb4932aec9f7cdde84d492 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22b8842ce70d4149995b07b71111972e : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fcf614bfd264eb4806860daa78059b4 : SUCCESS in 12m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b41c585796844d990422d8e30bfca11 : SUCCESS in 13m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07e94472a62049d386f87c533b16678d : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18672c6ee7c844c9b54ef2d38b6df38d : SUCCESS in 16m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/1fe11c5ee6ed4b0c82f89cf14d5ebe22 : SUCCESS in 1h 02m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25e4ed13b5894ffcba9c230f957af713 : SUCCESS in 1h 24m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9091ced327f64c2dbc55bdb89db7acbf : SUCCESS in 15m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/690d183d3bee450b9930e507bb0e82d1 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7152aef80df0441b93c8b6dcc2dc5c88 : FAILURE in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f04b49c76fc642d384693e0d92a01d08 : FAILURE in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/688b2b7076284d61a6780eabbd07dd65 : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0b529d5c62d402e9e6ec5468407a5f7 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6db87ab255084adc807401c5364e28fe : POST_FAILURE in 1h 17m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/733ef2ed00304a2884329816dc4ec8ec : SUCCESS in 1h 15m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fae91b1b4df34cf1af67f6053c0a0f1c : SUCCESS in 1h 18m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4e810604edb34ba1989475f2c0733e0a : SUCCESS in 1h 24m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e30ba71d148c413696707648fdabb338 : FAILURE in 1h 05m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7743bee9cb4948a8a6b04b71495b2384 : SUCCESS in 1h 14m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/634dd8e3a31b4b77b9df8c2cb6a061e4 : SUCCESS in 1h 59m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/936ef93a01ce4b479fad40fd1692f529 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":4},{"id":"f1325306de3e5c5f9be211640eb25f36f7eb697d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-11 17:52:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/789603/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/801e1f2/ : SUCCESS in 2h 00m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/789603/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6fb80ce/ : SUCCESS in 1h 50m 06s","accounts_in_message":[],"_revision_number":4},{"id":"a7a55f131908f9126574f5329e0434e1130d6bd0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 18:02:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/4/check/iscsi-driver-primera/a42ad2d/a42ad2d.html : FAILURE in 1h 10m 07s","accounts_in_message":[],"_revision_number":4},{"id":"06bb0e63eab9832d04b3a2c9f159c36dab173f11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-11 18:10:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/789603/4/check/cinder-storpool-tempest/65ece88/ : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":4},{"id":"5ea587772a0ebd9c4713483bd3e5e2edbded55df","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-11 18:16:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/4/check/dsvm-tempest-cisco-zonemanager-job/f803f02/ : SUCCESS in 1h 14m 19s","accounts_in_message":[],"_revision_number":4},{"id":"29220d504aa3707fc1b17b8152b62aa183c5a7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-11 18:54:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"151e6d9a78e0cf42aefdff82159144a1b3f83493","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 19:20:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5899585209a94429a6e36b24dc6135b5 : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fc4dcb9471d4ddc87054a13cce878d8 : SUCCESS in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3f81f9268e8774b14cd3d04578f09b8950a5dbdc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 19:22:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/789603/5/check/iscsi-driver-nimble/c85034b/c85034b.html : FAILURE in 25m 52s","accounts_in_message":[],"_revision_number":5},{"id":"acd97f72c96d4636bb080479802121453922071b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-11 19:49:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/789603/5/check/cinder-storpool-tempest/6026b6b/ : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":5},{"id":"91e6ca096e6b34abb686c6e790c5ac03d7bd81a8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-11 20:00:02.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-3/ns5_nfs : SUCCESS in 4h 43m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-3/ns5_iscsi : SUCCESS in 4h 32m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"5fc47531657a53dfd671420b195ef038481bab9c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-05-11 20:00:17.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"04ecc17ebddcd72a3341f89c3ff5d0e91fb37599","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-05-11 20:00:18.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"dc93cb8640212110aac5625e1fc1f8e95c454131","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-11 20:01:15.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/5/iscsi/ : SUCCESS in 1h 05m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/5/smb/ : FAILURE in 50m 08s","accounts_in_message":[],"_revision_number":5},{"id":"813adfc7e06d9874872334696921230d12b01019","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 20:06:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/5/check/iscsi-driver-primera/771b7de/771b7de.html : FAILURE in 1h 09m 54s","accounts_in_message":[],"_revision_number":5},{"id":"09789933773772a2d3df245c5bbf31784ea4136f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-11 20:12:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/5/check/dsvm-tempest-cisco-zonemanager-job/c9117cb/ : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":5},{"id":"7c2fdecea65d387b73520732d2091150eb34b268","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-11 20:37:46.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/5/check/tempest-dsvm-full-focal-py3/cfd799b/ : SUCCESS in 1h 31m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/5/check/devstack-plugin-ceph-dsvm-focal-py3/cbf2e87/ : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":5},{"id":"95f8f764e4a7a7ab870ec85388e4319919392960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 21:09:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ddf2729936c47f68760933467b2a90c : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd6213ef89ee46189b99c8cf03367db7 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5711f0d0bc2848c6b51e58668a0133b4 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a7eae6d3f38488eaa31a9bf1fa32d95 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aecd104c15c4f7389beed3e8685d735 : SUCCESS in 12m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9485541436bd451db4eff44c3f156775 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22148db036764b429b10d58a2166e4b2 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/a8d5d97081774019bf0bb97c1981d9d8 : SUCCESS in 1h 08m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad71482f3b784ab0b1026bcb65ae9906 : SUCCESS in 1h 05m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e307b17481084c7f9dff2d13a6d6f168 : SUCCESS in 14m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d0af12542154368addd7f0fbd859704 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53c1a809f0f74da6809b4df217dd3828 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af36484bb2854d96aab5d16fcf6cd372 : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4b931912ba2245abae2ed1b3d84fb0c9 : SUCCESS in 41m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad418a54e66e4871bfae07eaabd936f0 : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f85c8e98e4b74314ab5cf642608aeed4 : SUCCESS in 1h 13m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d77abc2810f41f28778dd1307781028 : SUCCESS in 1h 24m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/285d63bdfcf44c7f88331963fcfa863f : SUCCESS in 59m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2edd981fd5fa4e86b00fa026a9330552 : SUCCESS in 1h 18m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2161594fc91c47e2884558d0dffd44a9 : SUCCESS in 52m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0aa7f0d31fb14afaae0e212caa795a37 : SUCCESS in 1h 17m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a759e8958c14be9b0a031382ac902dd : SUCCESS in 1h 59m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d262e67cc5914629ac13898fde4c4abd : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":5},{"id":"b2607e06ab23d907ee4cc28ec977fd3b3cab881e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-11 22:00:50.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/789603/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/495573b/ : TIMED_OUT in 3h 04m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/789603/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b85b42c/ : TIMED_OUT in 3h 04m 15s","accounts_in_message":[],"_revision_number":5},{"id":"85664c09fe8b2686f9d360199e14b722aa630d5c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-11 22:06:51.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/789603/5/upstream-check/cinder-cDOT-iSCSI/b8a70ec/ : FAILURE in 36m 44s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/789603/5/upstream-check/cinder-cDOT-NFS/966e367/ : SUCCESS in 36m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/789603/5/upstream-check/cinder-cDOT-FCP/24ba113/ : SUCCESS in 56m 18s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"1cc7773cc3369fc6d7085e97f8f44cd233c475db","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-11 22:32:44.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-4/ns5_nfs : SUCCESS in 5h 1m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-4/ns5_iscsi : SUCCESS in 4h 44m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"ed0621baf4e0f0c92ea808ea7ceacf6f59a87829","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-11 22:55:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/789603/5/upstream-solidfire-check/cinder-solidfire-iSCSI/1dae1a4/ : FAILURE in 1h 44m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"63b1a2749a6d244e3d4c6e58911391acbe6b73eb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-11 23:08:17.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/789603/5/check/DellEMC_PowerStore_iSCSI/68547c4/DellEMC_PowerStore_iSCSI/79 : SUCCESS in 49m 26s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/789603/5/check/DellEMC_PowerStore_FC/d17e3a4/DellEMC_PowerStore_FC/80 : SUCCESS in 47m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"c59c50823dce93dd289fd2d92b36a2441562f509","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-11 23:55:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ba6b25d/ba6b25d.html : SUCCESS in 2h 19m 34s","accounts_in_message":[],"_revision_number":5},{"id":"721604fd32078db9327e0f85c6804edd374c5da6","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-12 00:51:08.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/789603-5/Seagate-iSCSI : SUCCESS in 45:04.65\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/789603-5/Seagate-FC : SUCCESS in 38:14.19","accounts_in_message":[],"_revision_number":5},{"id":"d6495cab5846936de0b57cccbbc1b5c1260e9c6a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-12 00:51:48.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-5/ns5_nfs : SUCCESS in 4h 50m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-5/ns5_iscsi : SUCCESS in 4h 48m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"f047edf13fa42006939e4585aba8b7dc20d311c4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-12 00:56:09.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/03/789603/5/check/DellEMC_PowerFlex/c6030c7/DellEMC_PowerFlex/87 : SUCCESS in 1h 19m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e6d6e4534753ea513c10dbe7de60cf66db17244","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-12 03:37:29.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/5/check/DellEMC_XtremIO_FC/0fe6eaa/DellEMC_XtremIO_FC/147 : SUCCESS in 2h 07m 53s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/5/check/DellEMC_XtremIO_iSCSI/a3d741b/DellEMC_XtremIO_iSCSI/148 : SUCCESS in 1h 53m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"d649ab6d9635584494bb7a44a011429980c47b4b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-12 04:46:31.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/03/789603/5/check/DellEMC_SC_FC/d39934b/DellEMC_SC_FC/95 : SUCCESS in 2h 16m 00s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/03/789603/5/check/DellEMC_SC_iSCSI/9ec7abc/DellEMC_SC_iSCSI/92 : SUCCESS in 2h 26m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"a8669d819e8499c9de5ff4b02b8cc09fcd929376","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-12 04:50:44.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/5/check/DellEMC_Unity_FC/0426564/DellEMC_Unity_FC/163 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/5/check/DellEMC_Unity_iSCSI/4ab3d08/DellEMC_Unity_iSCSI/149 : SUCCESS in 1h 29m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b0ccd92c988de76b6ba41a226250523a7e18abcf","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-12 05:06:51.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/789603-5/Lenovo-iSCSI : SUCCESS in 47:35.76\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/789603-5/Lenovo-FC : SUCCESS in 43:23.98","accounts_in_message":[],"_revision_number":5},{"id":"f2e0aec140fc787423926f108df89d98079d87c5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-12 13:17:28.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1\n\nThis needs to also validate specs on encryption-type-update, and not just create.","accounts_in_message":[],"_revision_number":5},{"id":"1c55767d33a63dd42df65fade91a59a661d1f672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-12 14:41:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"07d3d75d7b921e06bb7e2d92b9666e508ff6d7f7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-12 14:41:44.000000000","message":"Patch Set 6: Workflow-1\n\nNeeds tests added","accounts_in_message":[],"_revision_number":6},{"id":"fc32d38db02c8877d639bc6e45a2a3494c9c6dd9","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-12 14:42:52.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/789603-5/HPMSA-iSCSI : FAILURE in 1:26:00\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/789603-5/HPMSA-FC : SUCCESS in 32:49.63","accounts_in_message":[],"_revision_number":5},{"id":"3c38e39dede7b4355475ab444dcfeeab06902f65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:02:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0ac61050af04ef1a0b15b486443a648 : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d998bd491fdb46d4af1d06806203041d : SUCCESS in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7f9efb7647a20b20bb7a765e3a7abb618b0fd3be","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-12 15:09:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/789603/6/check/iscsi-driver-nimble/3e262b7/3e262b7.html : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":6},{"id":"6c17a07eedea6d781301631e8b41bed7694b4613","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-12 15:50:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/6/iscsi/ : SUCCESS in 59m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/6/smb/ : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":6},{"id":"b9d17cce52b4f7e9950fb761fd27935b9be4129e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-12 15:56:45.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/6/check/iscsi-driver-primera/a8b2a84/a8b2a84.html : FAILURE in 1h 10m 06s","accounts_in_message":[],"_revision_number":6},{"id":"f9ce00a2cc4bd2a3873cdfa1dc5119773881d327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 16:20:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e30b6d96a81421687c7bf9aa3e3a626 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cb065c643834ff8a57fff0043879b75 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b752ccdc2afd469f92a87776cc9667d1 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/409d92f781b14f44bfd538487c1dc8bf : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b20c0fea09784165adf4ba398db88dec : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07bc8f209b944b50b8fc3f3c26cf300c : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/3c2e457e23854fb187eb3543acd65ba9 : SUCCESS in 1h 02m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/397ce1aa51584b6f9401a4aab3ea8d7f : SUCCESS in 1h 10m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/81297fbf48e5430e9f4d098c3ae05fb8 : SUCCESS in 13m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d669017005a2410c8ea4243df22958a8 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e646a58afa645019a8cd906ddad1b72 : FAILURE in 6m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/801cfbb4430543c48c81d666056da1fc : FAILURE in 8m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a6feb2011fb34a94acadaa24d449412a : SUCCESS in 31m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78773246e8174dbb83c908f8fde986ee : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/31a059583e9a4d0e90df2ff8c5a5dd51 : SUCCESS in 1h 28m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80285cb6286b408994bcac60e4878577 : SUCCESS in 1h 18m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57ba8cb7c48642f8afdf4df1ec691b9b : SUCCESS in 1h 33m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8e953de793f54906b5978faef443a79e : SUCCESS in 51m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88b8ce955dde4b3d9041d342f5b316c7 : SUCCESS in 1h 05m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/01ae1f6105d84fb7b2973178da21b2ed : SUCCESS in 1h 00m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df46a97fbe264daca63052e1aeb0d65b : SUCCESS in 1h 33m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/502fce1675f64129947abe4a37f584b0 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":6},{"id":"71c851638b99779d59dd55097f82b79d5a931ae3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-12 16:23:48.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/6/check/DellEMC_Unity_FC/f6c8bcd/DellEMC_Unity_FC/167 : FAILURE in 1s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/6/check/DellEMC_Unity_iSCSI/06d1115/DellEMC_Unity_iSCSI/153 : SUCCESS in 1h 41m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"9a2acb885faf6e5b1faa86bd01393c0022c738f8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-12 16:24:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/6/check/dsvm-tempest-cisco-zonemanager-job/743ab6b/ : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":6},{"id":"8c52f69b278e6c5ff2008a29592fe558c644e909","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-12 16:34:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/789603/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f717714/ : SUCCESS in 1h 44m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/789603/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6381bb0/ : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":6},{"id":"dee45949d9475bb7e471ff56e81f766c1d2abb56","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-12 16:51:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3909251/3909251.html : SUCCESS in 2h 08m 18s","accounts_in_message":[],"_revision_number":6},{"id":"e7acd451a6e8bd84efb8649c9be0e7c5b20539d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-12 16:56:28.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/6/check/tempest-dsvm-full-focal-py3/9bbc308/ : SUCCESS in 2h 05m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/6/check/devstack-plugin-ceph-dsvm-focal-py3/cf94b78/ : SUCCESS in 1h 47m 36s","accounts_in_message":[],"_revision_number":6},{"id":"d635e224554ab7e2fe85203ca2613ae5062c3cb7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-05-12 16:56:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/03/789603/6/check/dsvm-tempest-zadara-driver/1607f8c : SUCCESS in 1h 08m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b222535eeaa04eb28a08b73485f82af0a82d7515","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-12 17:07:30.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/6/check/DellEMC_XtremIO_FC/67c6b46/DellEMC_XtremIO_FC/151 : SUCCESS in 2h 25m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/6/check/DellEMC_XtremIO_iSCSI/13baf60/DellEMC_XtremIO_iSCSI/152 : SUCCESS in 2h 22m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"df4fe06c63814ae68f1ce6720b3f219b61b2729e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-12 17:14:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/789603/6/check/cinder-storpool-tempest/6b56428/ : POST_FAILURE in 2h 30m 47s","accounts_in_message":[],"_revision_number":6},{"id":"29060ff063bd34c46c45a2355c91b4b1bd39d74e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-12 19:23:55.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-6/ns5_nfs : SUCCESS in 4h 41m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-6/ns5_iscsi : SUCCESS in 4h 28m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"c40b14186ffa8e9a58469fa028341439685521f8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-05-13 03:58:24.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/6/check-not-vote/ext-cinder-zuul/7f0d828 : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":6},{"id":"bc262681852f01d6f073d54f84e4f3bfd1049d0b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-13 08:13:02.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/6/20210513161236/NEC-ISCSI : FAILURE in 59m 08s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/6/20210513161236/NEC-FC : FAILURE in 1h 00m 23s","accounts_in_message":[],"_revision_number":6},{"id":"2724123da62f0a14b5d4c9dbb302d02661dd31e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-07 14:02:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1a3e369abf111d5f9b5e191d28d6c058dd2580e5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-07 14:03:01.000000000","message":"Patch Set 7: Workflow-1\n\nNeeds tests work","accounts_in_message":[],"_revision_number":7},{"id":"e9f1f3662a04aa53ed00665a8aa504c394ad79da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:27:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c01a9ce405240078c569d08460788b1 : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19bf0b9748914410b7a0a048b181e379 : SUCCESS in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7110ef3685d4637a902b4378dde4a54b772b4fb5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-07 14:39:47.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":7},{"id":"df17248c41d3877a03f4509944a6ea6cd02c0cce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-07 14:43:02.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/36cab1d/36cab1d.html : FAILURE in 25m 27s","accounts_in_message":[],"_revision_number":7},{"id":"10b059f8adb09cc28dd56386931d65ffad4d6fdc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-07 15:00:55.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/7/check/iscsi-driver-primera/b33d853/b33d853.html : FAILURE in 49m 30s","accounts_in_message":[],"_revision_number":7},{"id":"574198d09fbfe062f316fb6e9e52367c633021c1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-07 15:04:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/7/iscsi/ : SUCCESS in 1h 00m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/7/smb/ : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":7},{"id":"e194747e2a9d9d1b93ca599449744aaf4e3f3631","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-07 15:05:45.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/7/check/dsvm-tempest-cisco-zonemanager-job/b00d543/ : FAILURE in 56m 44s","accounts_in_message":[],"_revision_number":7},{"id":"117f45510027e596d84f0962629f364f2aa87e8c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-07 16:15:26.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/7/check/tempest-dsvm-full-focal-py3/5245ce9/ : SUCCESS in 2h 06m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/7/check/devstack-plugin-ceph-dsvm-focal-py3/4b0a034/ : FAILURE in 46m 49s","accounts_in_message":[],"_revision_number":7},{"id":"01203a73693f7a3b4915dfa193f0679e3325bb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:23:42.000000000","message":"Patch Set 7: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cbafbff3042418f86c7bdd72f848a1b : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6a4be7e1bb44cb98b7b98a02014784 : FAILURE in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae4702cebf2a450099addb76f60dc2bf : SUCCESS in 11m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a714662d6354dacbfde3e501f18b689 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6dc9b8658bf4118ac3570df2bc3a77c : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ac0d3c14d549b58dcfd65dfd14b34b : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/5a42f02369dd40e2a5978af50f3a93fe : SUCCESS in 53m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/824938c40ea84a8387644b74052f509c : SUCCESS in 1h 08m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5b26054159e04636b6d9a0d758f3c49a : SUCCESS in 18m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46896b205d5f4f8f99e0f00acf703610 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6487864cac0645ee8659100c97485bdd : FAILURE in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7edabace915d48a987c3f99f7b4bf04c : FAILURE in 8m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0fc21ce2ee56406d9c191e38f110cd6d : FAILURE in 37m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/320443ac822548589435a14ce86a6d3d : SUCCESS in 7m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5500c085ab6b486aaf31c88358ec2d2c : SUCCESS in 1h 20m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9ec1953f51ff42fcb51170363bf33dc6 : SUCCESS in 1h 52m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/16d038bd66d14a0480dd3945f0fdd8e1 : FAILURE in 1h 53m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40987c388c5a4d969dae6170afb3f751 : SUCCESS in 1h 18m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/233b219880414129bc75840606c3b7ef : SUCCESS in 1h 24m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/211142fc124b408ea2710210900d173d : SUCCESS in 1h 30m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8535f5459e1540af870e7c2b3ddb817b : SUCCESS in 2h 11m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4853cdc738d448b998dbba280f678a7a : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":7},{"id":"be04b9d7ac18246f497d4d2e91fb88c492f0a8ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-07 16:43:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bd9ba44852c28aaacad2080fd4e7b88a35cbfe63","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-07 16:43:53.000000000","message":"Patch Set 8: Workflow-1\n\nNeeds tests work","accounts_in_message":[],"_revision_number":8},{"id":"ff649111314691f98da32cb6aac2e6d485792e4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 17:07:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/604722a7b3b04a0a8e700e240efce6b2 : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19294b9e8592484298d3f8161a03ef27 : SUCCESS in 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"286ddc342756582094bacd30e94ea499be840e4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-07 17:16:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":8},{"id":"5fe0f2de8bb52ba7a14c582cbbfa6144c57f6e34","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-07 17:27:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e80ed02/e80ed02.html : FAILURE in 38m 08s","accounts_in_message":[],"_revision_number":8},{"id":"0a7a9989d2e6756831035159c24834f2ad4d3f0a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-07 17:33:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/8/check/iscsi-driver-primera/4ea8ff9/4ea8ff9.html : FAILURE in 48m 32s","accounts_in_message":[],"_revision_number":8},{"id":"b3c0a23357ec0561101cf311352bb37e56cae477","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-07 17:35:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/8/iscsi/ : SUCCESS in 51m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/8/smb/ : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":8},{"id":"8e4e1dd562a449de1842997b557c5092ac09502d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-07 17:46:48.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/8/check/dsvm-tempest-cisco-zonemanager-job/3943ba4/ : FAILURE in 57m 36s","accounts_in_message":[],"_revision_number":8},{"id":"3501a717abf01b4a209eceb5948c7509a2a3a8c8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-07 18:37:56.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-7/ns5_nfs : SUCCESS in 4h 34m 16s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-7/ns5_iscsi : SUCCESS in 4h 34m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"6747019f5f82863ad5f85f7663daeb359a44912d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:55:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a654618f3a749859d5957fa7e4efabd : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b70d40bab2042d6a341d4efc814e4e0 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/226af6f986514b4b92aa3630b3b66fb6 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84e7296f8ecf4095be4f0f11f6ff7866 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dedfaf908ba4db2926febd00a8079d8 : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb76216d82f54154b48b4a154b23145f : SUCCESS in 15m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/dceafba253224913b5a5cb3704dcc9bf : SUCCESS in 58m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ea4ee4b4b72487280ad403e2b3dcdb8 : SUCCESS in 1h 14m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f0143f9ec006451abf4298f70fdb234a : SUCCESS in 14m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43b4c7a35df2432abe7981baf6406787 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fb451cac5f54c99857d71a5c25285db : FAILURE in 7m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ae7097f753a4b15bf45574fd9ae2cab : FAILURE in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c476f4f403e04edebda02117f6d69295 : FAILURE in 40m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/abcbb7e961de4370959bd9f4521dcdbc : SUCCESS in 7m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d398a6dbfab04f6ca9f3f6fc6bf89b23 : SUCCESS in 1h 07m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2e22b216f2404cc7b560bdb5b22cd94d : SUCCESS in 1h 36m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bb0eff0192c848feb95bc1045a154e31 : SUCCESS in 1h 57m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79a5973445014db0bb9a998b93a1c156 : SUCCESS in 1h 24m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f340a3d0ceb54745929f4c231d40e037 : SUCCESS in 1h 33m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7cf090549e74a2ca568cc532b1f76df : SUCCESS in 1h 27m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0a4b1d6d62c43b993a792308c793628 : SUCCESS in 2h 09m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c5c3faa159a4fa5a308d99140061888 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":8},{"id":"346b48e8c1a4638aaab0bda1348aed69bbf21e69","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-07 19:08:24.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/8/check/tempest-dsvm-full-focal-py3/781d5f4/ : SUCCESS in 1h 35m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/8/check/devstack-plugin-ceph-dsvm-focal-py3/ea958d8/ : FAILURE in 59m 10s","accounts_in_message":[],"_revision_number":8},{"id":"93578c8593cd5954210918129bbc01a03ab42799","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-07 21:21:25.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-8/ns5_nfs : SUCCESS in 4h 33m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-8/ns5_iscsi : SUCCESS in 4h 37m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"c39b4398d97fb7bb1acb680873aa2fa8f9f3fb0b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 23:15:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/8/20210608053200/NEC-FC : SUCCESS in 2h 37m 07s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/8/20210608053200/NEC-ISCSI : SUCCESS in 2h 43m 18s","accounts_in_message":[],"_revision_number":8},{"id":"be6e4f95a7db0a1a847c3f8487789f910d692473","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-08 00:41:17.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/8/check/DellEMC_Unity_FC/027d700/DellEMC_Unity_FC/253 : SUCCESS in 1h 34m 30s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/8/check/DellEMC_Unity_iSCSI/a8ca933/DellEMC_Unity_iSCSI/238 : SUCCESS in 1h 52m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"1b63ed6806230e38c6e4d6d62759cfb2cc9c8eef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-08 01:00:01.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/8/check-not-vote/ext-cinder-zuul/3d34611 : FAILURE in 16m 20s","accounts_in_message":[],"_revision_number":8},{"id":"80b152afdb75a0d7981b8d26872001df3d91836d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-08 02:06:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/8/check/DellEMC_XtremIO_FC/e6a1ddb/DellEMC_XtremIO_FC/242 : FAILURE in 1h 33m 40s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/8/check/DellEMC_XtremIO_iSCSI/9254ae0/DellEMC_XtremIO_iSCSI/246 : SUCCESS in 1h 49m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"854b0a3f55bd5b246a8af515f03f73432432044f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-09 02:02:27.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/03/789603/8/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":8},{"id":"076ed9aa96decbdeaa2ca75b8195f15f4b8d4ddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-09 14:24:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e56c146fba961d6a7502a43086b41eb669b3c8b4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-09 14:25:08.000000000","message":"Patch Set 9: Workflow-1\n\nNeeds additional unit tests","accounts_in_message":[],"_revision_number":9},{"id":"a6c9e8f1a51752e3a7802c110d7087b28434721e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-09 14:36:33.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/9/check-not-vote/ext-cinder-zuul/24e0d94 : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":9},{"id":"9140454ebf691d41bdcbd02641bcd7f512cfe58b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:46:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a469875b2c554b98b8f5f8e0040104c2 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96881e6efd684d5a9f43bba3526ab76d : SUCCESS in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d74e7f85dc80b3ed0c850b589e23b57f9b2062a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-09 14:57:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":9},{"id":"1588ff6c8900e8e3d5e06a39306d7c54dd190734","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-09 15:16:26.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/dc8ea92/dc8ea92.html : FAILURE in 49m 48s","accounts_in_message":[],"_revision_number":9},{"id":"85903ae5bd9ab77e8985e6f6d3d47f84ba57f448","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-09 15:19:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/9/iscsi/ : SUCCESS in 54m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/9/smb/ : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":9},{"id":"8840b153f79c4a8f0c1afacc23969c8c8730bfba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-09 15:30:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/9/check/dsvm-tempest-cisco-zonemanager-job/8698409/ : FAILURE in 58m 56s","accounts_in_message":[],"_revision_number":9},{"id":"49b456df46953fb5e191dafa471e9ca8ec422652","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-09 15:30:53.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/9/check/iscsi-driver-primera/952bdfc/952bdfc.html : FAILURE in 1h 04m 14s","accounts_in_message":[],"_revision_number":9},{"id":"3f15b83491076aa7d016243126ebb6f9f0b8df4d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-09 15:34:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/03/789603/9/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":9},{"id":"c7b13e1116f5cf4876bfb73e74a71bb63d68f070","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-09 16:14:24.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/9/check/tempest-dsvm-full-focal-py3/5ae7a2f/ : SUCCESS in 1h 40m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/9/check/devstack-plugin-ceph-dsvm-focal-py3/429452b/ : FAILURE in 45m 23s","accounts_in_message":[],"_revision_number":9},{"id":"a2d994a48774bb4a4b15fbd132d868bf34b06e15","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-09 16:20:10.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_Unity_FC/eb71863/DellEMC_Unity_FC/265 : SUCCESS in 1h 45m 13s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_Unity_iSCSI/740a251/DellEMC_Unity_iSCSI/249 : SUCCESS in 1h 54m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"1f74d9e0cf6d0d89440728cbcf55e8ac99492633","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 16:31:30.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_XtremIO_FC/c8a42c3/DellEMC_XtremIO_FC/255 : SUCCESS in 1h 37m 29s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_XtremIO_iSCSI/3ce29a6/DellEMC_XtremIO_iSCSI/258 : SUCCESS in 2h 05m 59s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"e5ea8a050952151383f9d5d9f59003c07ab31fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:40:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/986bcf56cd004b66ad5a25bfc09efc53 : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f857c09f2dbd424e892c63b911721571 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076041f3cbfc49f4b5ccb7532a5dbf01 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c7e666cb0fe42caa4e8db89f187790a : SUCCESS in 13m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c724251c5a4412783c9ba24e55ea044 : SUCCESS in 12m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43157c2465c34ca180d5f2f9f7d0a641 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15714051b3b5439292f6b0add244a3f5 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/9fb51b59570c48cb8d8b851646a9bbbd : SUCCESS in 1h 02m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e20579c4b0584365a3c192978fdadd0d : SUCCESS in 1h 10m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac3a8fb7e8c24cc48e41056838e85237 : SUCCESS in 14m 07s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff9367070512461f9688747788a1404a : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/587ddc7885db483c80fd9783b7d9cd39 : SUCCESS in 6m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf8557d3a7194cb2a45533007b5d3841 : SUCCESS in 7m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f45fe1310974cfa980cd8191bd02628 : FAILURE in 33m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/31d2683b5c1e408ab5ba7ca2c6c02ffc : SUCCESS in 7m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd68df0605364343b4f09f870c18f9cc : SUCCESS in 1h 17m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b81d7d7825134217a8382397c7c8631f : SUCCESS in 1h 43m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f0e7b9ea47746deb81766b1e156c950 : SUCCESS in 1h 31m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a8fe244868d467b850b3141c3111ddb : SUCCESS in 1h 14m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/45d07adff8a84ec687e626c080e19f0d : SUCCESS in 1h 19m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c96edd29e694d23806280cf60eded6e : SUCCESS in 1h 18m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b173f26e990c493883b436ce805f62db : SUCCESS in 2h 10m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/006c6175c4c24f0e8b9f0ea1a1051704 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":9},{"id":"2535276f352026a3de3da1e5c166f595062a1197","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 17:05:41.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/9/20210609232428/NEC-FC : SUCCESS in 2h 33m 42s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/9/20210609232428/NEC-ISCSI : SUCCESS in 2h 41m 10s","accounts_in_message":[],"_revision_number":9},{"id":"b17ccd7f34296689d48f1d8abd2c952e5e2cc15d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-09 17:20:32.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/789603/9/upstream-check/cinder-cDOT-iSCSI/817b7ff/ : SUCCESS in 38m 29s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/789603/9/upstream-check/cinder-cDOT-NFS/c6e6bbb/ : SUCCESS in 38m 24s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/789603/9/upstream-check/cinder-cDOT-FCP/b9fca92/ : FAILURE in 9m 07s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"b1b2021181b5b5b0bf91b88e2a21c1c975f3623a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-09 17:48:01.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/789603/9/upstream-solidfire-check/cinder-solidfire-iSCSI/bcbd211/ : SUCCESS in 1h 05m 40s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"e1c773a4f24078daef0fa044a005c1e42e83dd93","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-09 18:20:19.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_PowerStore_iSCSI/454170a/DellEMC_PowerStore_iSCSI/138 : FAILURE in 1h 09m 43s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_PowerStore_FC/c2dd45f/DellEMC_PowerStore_FC/139 : SUCCESS in 1h 09m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"7eea4e0ff27b905755a939126dafeeb06c80da2c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-09 18:40:36.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-9/ns5_nfs : SUCCESS in 4h 12m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-9/ns5_iscsi : SUCCESS in 4h 15m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"c34e99706240d41bb3fc1601665111a7f8b0c1f8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 20:07:28.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/03/789603/9/check/DellEMC_PowerFlex/6c6eb5b/DellEMC_PowerFlex/159 : SUCCESS in 2h 04m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"597afde5409f82279948109c425a50a94e6bdd12","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-09 20:39:24.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_Unity_FC/3e07344/DellEMC_Unity_FC/267 : SUCCESS in 1h 46m 52s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_Unity_iSCSI/f0f5614/DellEMC_Unity_iSCSI/251 : SUCCESS in 2h 06m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"e8e58ab81d048c61ba61dbbad02d7e6d665dd46d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 21:13:53.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_XtremIO_FC/350ba62/DellEMC_XtremIO_FC/257 : FAILURE in 1h 46m 52s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_XtremIO_iSCSI/73362ef/DellEMC_XtremIO_iSCSI/260 : SUCCESS in 2h 20m 15s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"189caca1754ce9a1dd6d1c2931f54c331565706c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-06-09 21:34:29.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_SC_FC/02c8b9f/DellEMC_SC_FC/158 : FAILURE in 1h 06m 49s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_SC_iSCSI/33e798e/DellEMC_SC_iSCSI/155 : FAILURE in 2h 53m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"1ea29652a33b170d14035a75ad36fdeabdb488e1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 23:08:23.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/03/789603/9/check/DellEMC_VNX_iSCSI/98e9995/DellEMC_VNX_iSCSI/153 : SUCCESS in 3h 28m 18s\n- DellEMC_VNX_FC http://publiclogs.emc.com/03/789603/9/check/DellEMC_VNX_FC/de2a3fe/DellEMC_VNX_FC/158 : FAILURE in 2h 09m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"16ff01f513f48d561cac32b8ff17dd4e54696f15","author":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"date":"2021-06-10 03:09:09.000000000","message":"Patch Set 9:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":9},{"id":"1d189a365009c663d5d062346d23ba50d61476b2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-10 03:50:41.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/9/check-not-vote/ext-cinder-zuul/52971b8 : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":9},{"id":"6400f537530b5f39bd9d897a854ba8d9a89d3f27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-10 16:29:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f10dd86dbad599cad42fcecabbce467dc33a7a19","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-10 16:30:38.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/10/check-not-vote/ext-cinder-zuul/f6321d9 : FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"d9847e26ea3c3fbcff1baf0106f9b0c80a7b9654","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-10 17:02:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":10},{"id":"57edcd1dda7a48d7cbb6a0f1806ebc206f8e51f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:10:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e0fa778e40f480d913762dfb5a89c86 : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19582ccfc0284d0eb565ac1239494366 : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c9dfd2996d675d620570169aeee399c78563deb5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-10 17:19:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/10/check/iscsi-driver-primera/6cb061e/6cb061e.html : FAILURE in 48m 24s","accounts_in_message":[],"_revision_number":10},{"id":"0be48071bded89984164a2209060522098a46e62","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-10 17:22:57.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9e633b4/9e633b4.html : FAILURE in 51m 14s","accounts_in_message":[],"_revision_number":10},{"id":"c61c81c198bcc6b95d8dceb11e502747a2c7beb1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-10 17:35:09.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/10/check/dsvm-tempest-cisco-zonemanager-job/23baf79/ : FAILURE in 59m 10s","accounts_in_message":[],"_revision_number":10},{"id":"7c39fb42bcdfd253301a004d9e68927fc8179504","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-10 17:40:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/10/iscsi/ : SUCCESS in 54m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/10/smb/ : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":10},{"id":"3892f1e0d1badf258bae17090d87796ce8b633c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:23:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6b3aad3fec3a46a99da27c897824c721 : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31a816f1802d48fd98e04da3932dab02 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf75fe336704e6c908ce9168ebdb5ca : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd79aea91b204e7d8b752c7feadd1aef : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9ec2db3de7241f3909e450ea3fda165 : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3917071a4efd4cc095c758abeb94ac3d : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea225626d854996a27fa844d072e0c3 : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5999339df95e4c3680cb849035ab07dd : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/755bd4d06fff4ed18d06600d560b805a : SUCCESS in 1h 10m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6205d35014b44813af6e8b0d21474b41 : SUCCESS in 14m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/edc0fdf203bb428ca698c004cd2c4f38 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74188a0d694843c3899e3caaef20525a : SUCCESS in 7m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/447fa40e7582444eaf441bd8b2bba176 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/00ffc0241e084a4d96558209190726c0 : FAILURE in 36m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/911929dd698f44be8729246252d07f89 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea3db3e7a02d4673a1f367c31e7a557d : SUCCESS in 1h 13m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a55829cfa12470c958bf9666b7d5c33 : SUCCESS in 1h 27m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/58cbe89234594d51a2a5c5ce3b93fc6a : FAILURE in 32m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c24e13f8548453484a68763c4beb643 : SUCCESS in 1h 22m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65c3b08fab28405aa20a40779c409ef5 : SUCCESS in 1h 22m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9339625b5fb34f1986f0f71eeac48cff : SUCCESS in 1h 22m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ff56e67de674f2e97647fc60f6bb224 : SUCCESS in 1h 50m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ade46944caf342d7861ecf5d215f7304 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":10},{"id":"d5d6fbec0dc8432d1291f30b9344cd7dfdf87977","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-10 18:30:08.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/10/check/tempest-dsvm-full-focal-py3/a5dfac6/ : SUCCESS in 1h 45m 41s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/789603/10/check/devstack-plugin-ceph-dsvm-focal-py3/41a4b99/ : FAILURE in 1h 40m 43s","accounts_in_message":[],"_revision_number":10},{"id":"a09da23c13a89f7cb113d122b7d1f04e1a1302c4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-10 22:35:10.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/10/check/DellEMC_XtremIO_FC/fc4e158/DellEMC_XtremIO_FC/268 : FAILURE in 1h 13m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/10/check/DellEMC_XtremIO_iSCSI/b16682d/DellEMC_XtremIO_iSCSI/271 : SUCCESS in 1h 51m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"d8255d91a0af4abdfb46a6d5719397335bee6b6f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-10 23:03:45.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/10/check/DellEMC_Unity_FC/7d3501e/DellEMC_Unity_FC/278 : FAILURE in 9m 26s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/10/check/DellEMC_Unity_iSCSI/d3cd20f/DellEMC_Unity_iSCSI/262 : SUCCESS in 2h 24m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"44eaf45ceaac250d2ee4175cd642abdbc16d01ef","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-11 01:26:59.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-10/ns5_nfs : FAILURE in 4h 34m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-10/ns5_iscsi : FAILURE in 4h 39m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"b8766134620870974f1f575cf506945dc73fccf0","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-16 12:56:45.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"1ae4ee9cc0b163305658771a72af5e211efdd740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:21:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/572db4d64e4a431eb08f45c1fb6b8bfa : SUCCESS in 15m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60ba5f4498a9411086f6aa704e6dd773 : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d3992148cbd6c1407dcc0d0a56378bcd0c6ad22b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-16 14:07:46.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/10/iscsi/ : FAILURE in 1h 10m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/10/smb/ : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":10},{"id":"3898a484c17338cf0bd64a90d41e36be13cf27ac","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-16 14:37:57.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/10/check/dsvm-tempest-cisco-zonemanager-job/8ef105f/ : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":10},{"id":"b84da3347fdf8dc2e47f2b41ae09c5b83938add0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-16 15:03:56.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/391ce423b1a0424d820cb4ffd0881345 : POST_FAILURE in 44m 42s","accounts_in_message":[],"_revision_number":10},{"id":"49e1c9f481f1ad30f4d53687ff72aedd698f4487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:44:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9fc3df4664784951b42b5e0684f13f10 : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f66a0eaf63e54ae597cd13ef98a40b21 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77b98e5a7ec4b30a341636454330f8d : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1883c188a5ba4373804b13ec0b3ccf13 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85cfbb34aecc46d1876d9f5372427011 : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a96d101d9884c479c1ddc8634177e66 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9ebebc99504f8d8da4880b3544b2a6 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9c42cb40db8f49ffb2e9c00b51109b57 : SUCCESS in 1h 00m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/073041fd9792418582768e0d0cced75e : SUCCESS in 1h 04m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28cfa48c6c664bd09b2852f1a912b7f4 : SUCCESS in 12m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb7c0aeaf4294a82af8cc3849c0936cd : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e255a9d50994a71a17a69a20c8315da : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45d3bdf507bb423ba8ada52e4c121d72 : SUCCESS in 8m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6cd00641a6ed486ba91eb1f8f44c5d70 : FAILURE in 32m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f8394a2e3cf449c83fb2db1eb81b5a2 : SUCCESS in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/546cc5c934734e48a55da0119ea408a2 : SUCCESS in 1h 19m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c0284a8d233440169e0dfcc7a8169ca7 : SUCCESS in 1h 38m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2943a5f8563e43ad9b42c317b04b55c2 : SUCCESS in 1h 38m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d9a6a9254d7a451c9e05db7c964ffe0e : SUCCESS in 1h 19m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d508e88e9658490fa0686558b5af8cba : SUCCESS in 1h 15m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/631c2e08d64e443197391881d4b6251a : SUCCESS in 1h 19m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf89c30d82184e909dd66a0ea0b8af46 : SUCCESS in 1h 48m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5164d29628b84b239e96a20374beb994 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":10},{"id":"4a1134abd5645679c0a3e0bbb75aff197183c70a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-16 16:48:20.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/789603/10/upstream-solidfire-check/cinder-solidfire-iSCSI/ddd08ad/ : SUCCESS in 1h 02m 15s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"f2496ea7af5efe293336f0c67128979b3f746880","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-16 20:28:12.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/789603/10/upstream-check/cinder-cDOT-iSCSI/c48dc0e/ : FAILURE in 1h 58m 59s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/789603/10/upstream-check/cinder-cDOT-NFS/f299274/ : SUCCESS in 33m 19s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/789603/10/upstream-check/cinder-cDOT-FCP/601db3c/ : SUCCESS in 58m 09s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"44d4844d0f2a0ae0384595b91c2215ee171e7889","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-16 21:49:43.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/789603/10/check/DellEMC_PowerStore_iSCSI/6f280d5/DellEMC_PowerStore_iSCSI/151 : SUCCESS in 1h 20m 46s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/789603/10/check/DellEMC_PowerStore_FC/51554a0/DellEMC_PowerStore_FC/152 : SUCCESS in 54m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"8dc8b347968ab6f2516ff59be865b8964a4e3e0e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-16 21:51:53.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)\n\nLooks good to me!","accounts_in_message":[],"_revision_number":10},{"id":"3738c8ba3f4b3540d5de0b2fa9bca6b7d6e1576c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-17 18:20:18.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/10/check/DellEMC_XtremIO_FC/66a32fa/DellEMC_XtremIO_FC/314 : SUCCESS in 1h 18m 37s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/10/check/DellEMC_XtremIO_iSCSI/cb9af25/DellEMC_XtremIO_iSCSI/317 : SUCCESS in 2h 14m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"23e5ba599e8946e9095522b8fe0b1e2b708d4c2f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-19 14:22:52.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/789603-10/Lenovo-iSCSI : FAILURE in 44:32.20\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/789603-10/Lenovo-FC : FAILURE in 42:19.69","accounts_in_message":[],"_revision_number":10},{"id":"e19510991ef9431bb4574cf0686463acdab61902","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-19 20:23:47.000000000","message":"Patch Set 10:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/789603-10/PVME-iSCSI : FAILURE in 52:25.36\n* PVME-FC http://os-logs.tristero.net/dell-ci/789603-10/PVME-FC : FAILURE in 51:40.67","accounts_in_message":[],"_revision_number":10},{"id":"8a7934a5632420f08ed12d3458af7643993a02b1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-06-21 05:17:03.000000000","message":"Patch Set 10:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/789603/10/fc : FAILURE in 0h 11m 17s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/789603/10/iscsi : FAILURE in 0h 16m 58s","accounts_in_message":[],"_revision_number":10},{"id":"58853d222b64c4978203dae08d71b74471da64a1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-23 17:12:06.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1908d996183eb1db418a31ef8457a30e063d25ba","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-07 20:07:06.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\nSmall problem with the commit message noted inline.  Code and tests look good to me.\n\nFor other reviewers: see https://review.opendev.org/c/openstack/tempest/+/791850 for discussion with the QA team about this change not requiring a new microversion.\n\nI think this could use a release note?  Maybe something along the lines of \"Cinder previously expected that an encryption spec would always contain \u0027key_size\u0027 and \u0027cipher\u0027 fields.  This expectation is now enforced when an encryption spec is created or updated.\"","accounts_in_message":[],"_revision_number":10},{"id":"abefffc4066925bdd3ceb069026044add4d13753","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-08 12:13:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e23a1353868e477898ddf426c225e9534cabc321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-08 12:22:10.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4f95c92d3a3af84e1a378f5d8c524bf8d1b17ba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 12:43:19.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/363f3ef0899e494ba8b303c293256a44 : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/702355a8d6c04692b07d418b92710626 : SUCCESS in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4fbacb4e8ccd310ee327a652f66b1be1e114e618","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-08 12:48:10.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"f2d12870d849612c5def429d86314fbf3544387f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-08 13:13:50.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/11/iscsi/ : SUCCESS in 50m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/11/smb/ : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":11},{"id":"fbfa15a4ad83de5470126ffab200de1abd024052","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 13:36:18.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/11/check/iscsi-driver-primera/1985838/1985838.html : FAILURE in 1h 05m 35s","accounts_in_message":[],"_revision_number":11},{"id":"2a2c3acf2f80751d03950de50f2d47cf4e69ad2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-08 14:01:44.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/789603/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4151593/ : SUCCESS in 1h 18m 37s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/789603/11/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cd7d360/ : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":11},{"id":"a85757b610b14dd6cc9d6bbd4cdf08856bd1d650","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-08 14:40:27.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/789603/11/check/dsvm-tempest-cisco-zonemanager-job/6177e42/ : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":11},{"id":"13f34663a163f15fca8216d7364ef05f477c3230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 14:41:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/026675d49a7144f4808613295b5b54d7 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1541d168754d426099dd25cb8852e915 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82444d47fc6b46d2be90c405d6268e15 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2b5915e6955403baeaa2a3e7e2a9719 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33fb1935e5eb48359ec503701fcc6bec : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88e6f4ab29234cf2a61d1a8657a2db4a : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aedf6c9ecf841f1856ee8cb97f18048 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/12dadd19d2a54be6bc9ca89238c70488 : SUCCESS in 1h 08m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2aea73544028494b9c35a6eabd29aad9 : SUCCESS in 1h 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43d7240de8064b5397111bfc99fa5c6c : SUCCESS in 7m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d1700252b0e4606a4f8e99b33c9a8ef : SUCCESS in 12m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/049deb4441a24386bb38259ae05ae4aa : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8440a4d3d2c4834a453e92d4482a6d2 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/46dbaa1f12ac4f37b22012017f37c437 : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c8be38b3a9b042ffbffb56a4c563bc59 : FAILURE in 38m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14a1572787534da88a2bc03ca49b0e73 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f82d9a618b534fd496748157e76b8cb5 : SUCCESS in 1h 17m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd6e150dde6440048dedc006d142882b : SUCCESS in 1h 36m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92d1911fb2f1409684efea64a51a61d6 : SUCCESS in 1h 28m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cc50c44f1f5447db9e44666b44fcd18 : SUCCESS in 1h 16m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0f24c3bbcca545dd9e88f5a29dcdb224 : SUCCESS in 1h 30m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/abf2cf33ee934dcdbf57159f73ba4816 : SUCCESS in 1h 14m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c30ba5665918431399fa3bce6e4c5f84 : SUCCESS in 2h 13m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/392d77614b174ebf98228298ed8c332d : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":11},{"id":"83d4f742d08519a24a25117c711d23fb20ee4fb1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-08 15:52:05.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/789603/11/upstream-check/cinder-cDOT-iSCSI/ff926bb/ : SUCCESS in 37m 18s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/789603/11/upstream-check/cinder-cDOT-NFS/48e6d3e/ : SUCCESS in 36m 45s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/789603/11/upstream-check/cinder-cDOT-FCP/3256944/ : SUCCESS in 1h 09m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"93c20c9e9419c2cb16cd61cfbb25365f72f96751","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-08 16:01:48.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/789603/11/check/DellEMC_PowerStore_iSCSI/9a8c53c/DellEMC_PowerStore_iSCSI/220 : SUCCESS in 52m 43s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/789603/11/check/DellEMC_PowerStore_FC/d950c15/DellEMC_PowerStore_FC/221 : SUCCESS in 54m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"768171ec9129c0524df5e0297292552623a445ca","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-08 16:27:40.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/789603/11/upstream-solidfire-check/cinder-solidfire-iSCSI/0c18e23/ : FAILURE in 1h 43m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"03da1778bb61d2acf155449ee355010c8ac530d7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-08 17:22:52.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/03/789603/11/ : FAILURE in 1h 20m 43s","accounts_in_message":[],"_revision_number":11},{"id":"dcbb618c1aa9badf0a8b6f0dca814d38aafb91f5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 17:34:56.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/03/789603/11/check/DellEMC_VNX_iSCSI/13df457/DellEMC_VNX_iSCSI/250 : FAILURE in 1h 14m 50s\n- DellEMC_VNX_FC http://publiclogs.emc.com/03/789603/11/check/DellEMC_VNX_FC/58a4710/DellEMC_VNX_FC/268 : FAILURE in 50m 49s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"9137abc8023f36bca596abd88776686704cf8eaf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 17:45:34.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/03/789603/11/check/DellEMC_PowerFlex/c2b1cb8/DellEMC_PowerFlex/288 : FAILURE in 1h 40m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"e46e14d42f1fe5614feb17654e3e10512f836db4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 17:50:30.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8668916/8668916.html : TIMED_OUT in 3h 01m 00s","accounts_in_message":[],"_revision_number":11},{"id":"fa0d41e1e2103b466f43ab3f80ee41ef67c22f1d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-08 18:18:59.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-11/ns5_nfs : FAILURE in 4h 30m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-11/ns5_iscsi : FAILURE in 4h 35m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"82bf0536f28791d554162bdfe1ac7dab2458542e","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-08 18:44:04.000000000","message":"Patch Set 11:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/789603-11/PVME-iSCSI : FAILURE in 9:24.43\n* PVME-FC http://os-logs.tristero.net/dell-ci/789603-11/PVME-FC : FAILURE in 47:04.92","accounts_in_message":[],"_revision_number":11},{"id":"f2ad3b776a1009fc47a19b7d7feb686d4aaae3d7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 19:42:31.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/11/check/DellEMC_XtremIO_FC/7739309/DellEMC_XtremIO_FC/468 : SUCCESS in 1h 22m 32s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/11/check/DellEMC_XtremIO_iSCSI/c29eec6/DellEMC_XtremIO_iSCSI/459 : SUCCESS in 1h 57m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"006f05e9078e41743cd858389b7e11ce21aaaf11","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-08 20:27:23.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/11/check/DellEMC_Unity_FC/77ae3b6/DellEMC_Unity_FC/400 : SUCCESS in 1h 30m 17s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/11/check/DellEMC_Unity_iSCSI/458fb0c/DellEMC_Unity_iSCSI/376 : SUCCESS in 1h 56m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"5f2bdb91890f9816e6c5b5b7f0c7df1bba8bfecd","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-08 20:49:47.000000000","message":"Patch Set 11:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/789603-11/Seagate-iSCSI : FAILURE in 43:43.40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/789603-11/Seagate-FC : FAILURE in 38:58.75","accounts_in_message":[],"_revision_number":11},{"id":"65f1cf3d2dd5929e0482441e5dac4083a492ec60","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-08 22:00:05.000000000","message":"Patch Set 11:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/789603/11/fc : FAILURE in 1h 10m 32s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/789603/11/iscsi : FAILURE in 1h 01m 15s","accounts_in_message":[],"_revision_number":11},{"id":"0ea2ec50653137f4d5b1c1a7b6ced42f3dd355e3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-08 22:06:05.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/789603-11/Lenovo-iSCSI : FAILURE in 47:37.79\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/789603-11/Lenovo-FC : FAILURE in 46:20.92","accounts_in_message":[],"_revision_number":11},{"id":"0e5e749e10ae13111c8e8d4374d7b214908f3a10","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-08 22:07:04.000000000","message":"Patch Set 11:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/789603-11/HPMSA-iSCSI : FAILURE in 59:32.35\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/789603-11/HPMSA-FC : FAILURE in 34:13.36","accounts_in_message":[],"_revision_number":11},{"id":"2b37a0b6a2cd1893580defbb6afa7aaab0c12017","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-08 23:33:26.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\nAll my concerns have been addressed.","accounts_in_message":[],"_revision_number":11},{"id":"ed14dec9db307fc1cb1cfa0ce8125f63b6bf3521","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-08 23:46:21.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/11/check-not-vote/ext-cinder-zuul/5327d35 : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":11},{"id":"165839c813c5897bca4c0c6891f2fa40d1c71fa8","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-07-09 12:00:30.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"88bc42c125079946f8477a73ce69779d79db396e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-09 14:33:09.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/11/20210709181845/NEC-FC : FAILURE in 5h 04m 05s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/11/20210709181845/NEC-ISCSI : FAILURE in 5h 14m 22s","accounts_in_message":[],"_revision_number":11},{"id":"ebf3ed482651cfdf406ee79b30711370262b0d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-15 15:01:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"79821233b7cab639ce27b875c9234facfe045da4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-15 15:02:21.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2aa97ab67a27ccfdd8dbe39cabab68a78933fde5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 15:23:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/700d4d05d69f497d8def58cb93c5496f : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6623632e4c346388d32a36b13aa129f : SUCCESS in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"172b173083a9f325b60a331a6c8cca209d3e16aa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-15 15:35:08.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"4d6d3acd796a076eea17843144bd829ec1327a57","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-15 15:36:06.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/12/check-not-vote/ext-cinder-zuul/22015e1 : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":12},{"id":"fc75484520ae1a09692a29d6525ec5a0a4884710","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-15 15:53:05.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/789603/12/iscsi/ : SUCCESS in 50m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/789603/12/smb/ : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":12},{"id":"05fe9eddd6196a140aa498eae3918e1de2d440b4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-15 16:04:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/03/789603/12/ : FAILURE in 48m 30s","accounts_in_message":[],"_revision_number":12},{"id":"d4468329df6dda23c7f5803340ab849606ab6601","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 16:06:37.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/789603/12/check/iscsi-driver-primera/3706f8b/3706f8b.html : FAILURE in 1h 02m 47s","accounts_in_message":[],"_revision_number":12},{"id":"0d257b75a65ba5563a8ffb694bd95d73cd3093a3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-15 16:10:24.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/12/check/DellEMC_Unity_FC/ee3eed3/DellEMC_Unity_FC/465 : FAILURE in 3m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/12/check/DellEMC_Unity_iSCSI/d987708/DellEMC_Unity_iSCSI/436 : FAILURE in 4m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"99e059efa815d86b43667a1b9d0f77f7632f471b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 16:23:34.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/12/check/DellEMC_XtremIO_FC/0581953/DellEMC_XtremIO_FC/529 : FAILURE in 1m 24s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/12/check/DellEMC_XtremIO_iSCSI/44e15a0/DellEMC_XtremIO_iSCSI/515 : FAILURE in 3m 30s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"f4fa70c19b91cf2e9c437fffeb64bdcead4c5857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:22:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dbb46e9f1ea349fd84553552ff42bec2 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0383af5942554fa3a20032cb065ff27d : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a03e06cd9794c22944ffd81a8d79bb9 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3309d35919994ec39236c9b1c64ada9a : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a555b35ce4384e6697eb9e5a52175904 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bc9a1f7eb24fb98b186f2620017f54 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/b3e0904a371e499e805311bf504cf791 : SUCCESS in 1h 01m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2eb01b36146042f9a2db4982b0dd834d : SUCCESS in 1h 25m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbce68446c824ceda1d77a7e52d9073c : SUCCESS in 6m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da5fb6a2d87847cd9a8c9010c6ca3709 : SUCCESS in 12m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e745fd088f5447e880c0a36b6cb8997f : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be8d636bef3e4b0f8fa7ad946ea2d0ca : SUCCESS in 6m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3617cf625e7a4ab1a129c5dbffdb0b90 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e6560c4fcf24431921266c6e642ebd3 : FAILURE in 36m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d78dbe1a9a8c486097e9be973ad0606b : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20fa9323a578450cb643de534ea71d1f : SUCCESS in 1h 35m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4ba9f0051b9444eabde8fdf0b88b08f : SUCCESS in 1h 22m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef71a0bc0d6046e88b1404f4050bf4ac : SUCCESS in 1h 30m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b7309b38d0d64b8ab5420fbcbca4119b : SUCCESS in 1h 18m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0dbd21a51f4e420d9fb5ae6775b8399c : SUCCESS in 1h 16m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5fdb33f6cd4449f2bdf8cc537786616e : SUCCESS in 1h 21m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f9a4d811b744b47b9f89fe87b6656ad : SUCCESS in 2h 15m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6217f2c168334370922061bb20abfbd4 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":12},{"id":"ec0efdef54fe49d256d2cd779c87c8f61167fc60","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 17:27:27.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/789603/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9595581/9595581.html : SUCCESS in 2h 23m 54s","accounts_in_message":[],"_revision_number":12},{"id":"e733ebb59d94a3ad7aa8c9ae1598bd3aacac9ca3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-15 18:05:11.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/789603/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/96bb2d4/ : FAILURE in 2h 54m 02s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/789603/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ca59b84/ : FAILURE in 3h 02m 09s","accounts_in_message":[],"_revision_number":12},{"id":"6849589b6977070eafd5ecf83dddc62fb703c0d4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-15 18:05:27.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/789603/12/check/DellEMC_PowerStore_iSCSI/a78840e/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/789603/12/check/DellEMC_PowerStore_FC/dbcbcf0/DellEMC_PowerStore_FC/244 : SUCCESS in 41m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"528898c4a15edca350d5fff66db08af5247d71d6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-15 18:33:09.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/789603/12/upstream-check/cinder-cDOT-iSCSI/2e0510d/ : FAILURE in 52m 24s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/789603/12/upstream-check/cinder-cDOT-NFS/de26da2/ : SUCCESS in 36m 22s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/789603/12/upstream-check/cinder-cDOT-FCP/db00f88/ : FAILURE in 1h 09m 19s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"91f31b9ebab20c1f9e4bc55286ae411ff38f9845","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-15 18:33:42.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/789603/12/check/DellEMC_Unity_FC/e7edf70/DellEMC_Unity_FC/468 : FAILURE in 1h 10m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/789603/12/check/DellEMC_Unity_iSCSI/2471843/DellEMC_Unity_iSCSI/439 : FAILURE in 1h 08m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"549e3a48b752db99b3ab92dc4c68b67ed487d040","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 19:01:57.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/789603/12/check/DellEMC_XtremIO_FC/2a64326/DellEMC_XtremIO_FC/532 : SUCCESS in 1h 11m 40s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/789603/12/check/DellEMC_XtremIO_iSCSI/28b1cc7/DellEMC_XtremIO_iSCSI/518 : FAILURE in 1h 38m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"09e9ded1d15541b7eeddba8a8cde6e1482e301dc","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-15 19:44:05.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/789603/12/upstream-solidfire-check/cinder-solidfire-iSCSI/cce0abc/ : FAILURE in 2h 19m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"30287a5fe12c320fcc362bb2f16c5fa09bec2878","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-15 20:03:35.000000000","message":"Patch Set 12: Code-Review+2 Review-Priority+1\n\nStill looking good.","accounts_in_message":[],"_revision_number":12},{"id":"e5c66337dc75c5e4a7b53b0d7116a9024d67a3bc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-15 20:09:05.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-12/ns5_nfs : FAILURE in 4h 43m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-789603-12/ns5_iscsi : FAILURE in 4h 47m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"3be2e63425fbb647abdfcc58fb5d78596f493404","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-15 20:38:54.000000000","message":"Patch Set 12: Code-Review+1\n\n(4 comments)\n\nYou asked me about where the validation should happen, and I completely forgot until now.  Sorry about that.  See comments inline.","accounts_in_message":[],"_revision_number":12},{"id":"b91b198458c1cc7ec21616acfc943d06afee3d3e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-15 20:41:20.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"561c3e771832c5625d8900ef0ee58c13b72c9479","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-15 23:04:38.000000000","message":"Patch Set 12:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/789603-12/Seagate-iSCSI : FAILURE in 45:34.89\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/789603-12/Seagate-FC : FAILURE in 43:55.08","accounts_in_message":[],"_revision_number":12},{"id":"b320ce381d100f184fe9cab69ab7fc2a3a7d327f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-15 23:17:39.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/789603-12/HPMSA-iSCSI : FAILURE in 58:45.45\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/789603-12/HPMSA-FC : FAILURE in 32:19.66","accounts_in_message":[],"_revision_number":12},{"id":"57edfc5d5b8346d587056a89129244aee07dd7d9","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-16 03:59:29.000000000","message":"Patch Set 12:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/789603-12/PVME-iSCSI : FAILURE in 43:08.68\n* PVME-FC http://os-logs.tristero.net/dell-ci/789603-12/PVME-FC : FAILURE in 46:59.53","accounts_in_message":[],"_revision_number":12},{"id":"d1ddb61a3613bcc4c6a7feae529082b7193eaea4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-16 17:00:05.000000000","message":"Patch Set 12:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/789603/12/fc : FAILURE in 0h 48m 07s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/789603/12/iscsi : FAILURE in 1h 10m 39s","accounts_in_message":[],"_revision_number":12},{"id":"8c45bb0ed230ae4969b0373cc2aca3c5bfdeb71f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 18:44:10.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/03/789603/12/check/DellEMC_PowerMAX_FC/c08bb8a/DellEMC_PowerMAX_FC/262 : FAILURE in 5h 12m 46s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/03/789603/12/check/DellEMC_PowerMAX_iSCSI/4a9a704/DellEMC_PowerMAX_iSCSI/247 : FAILURE in 3m 50s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":12},{"id":"a5dce14a191751e4fed83b8c447daa3b45c559ff","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-18 06:55:24.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/12/20210718103130/NEC-ISCSI : FAILURE in 5h 19m 50s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/789603/12/20210718103130/NEC-FC : FAILURE in 5h 23m 50s","accounts_in_message":[],"_revision_number":12},{"id":"a95974231805a4cc16375ecc6174297c3870fd0b","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-07-19 02:23:05.000000000","message":"Patch Set 12:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":12},{"id":"1cace3303c691b69d525a12db4d280ffc34d4eec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-19 02:59:23.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/789603/789603/12/check-not-vote/ext-cinder-zuul/791a46d : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":12},{"id":"11a83b38664865c1a399190b824794a67653d938","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-08-10 06:25:15.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"0cf17f42a091d66b4214ca7b088d595f124c5d04","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-17 16:01:29.000000000","message":"Removed Review-Priority+1 by \u003cGERRIT_ACCOUNT_5314\u003e\n","accounts_in_message":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"_revision_number":12}],"current_revision_number":12,"current_revision":"5d7142711b2301e3aecba3205610062dc124d311","revisions":{"6113b8da3c119d9310373e264e198d86e37ae716":{"kind":"REWORK","_number":1,"created":"2021-05-04 16:21:36.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/1"}}},"commit":{"parents":[{"commit":"baeaaa1f51825f895bf0817b8c45b80378a7af7a","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baeaaa1f51825f895bf0817b8c45b80378a7af7a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 16:19:24.000000000","tz":0},"subject":"Validate cipher in encryption spec","message":"Validate cipher in encryption spec\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6113b8da3c119d9310373e264e198d86e37ae716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6113b8da3c119d9310373e264e198d86e37ae716"}]},"branch":"refs/heads/master"},"06c591555f4af599e2143211e261bcc096b1884d":{"kind":"REWORK","_number":2,"created":"2021-05-04 17:36:12.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/2"}}},"commit":{"parents":[{"commit":"4f4f76ff459968feb9939d2125ce35748f79ff3b","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f4f76ff459968feb9939d2125ce35748f79ff3b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 17:35:43.000000000","tz":-240},"subject":"Validate cipher in encryption spec","message":"Validate cipher in encryption spec\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06c591555f4af599e2143211e261bcc096b1884d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06c591555f4af599e2143211e261bcc096b1884d"}]},"branch":"refs/heads/master"},"0334b47c4039692e51af3c3929dd75bc86ee6b53":{"kind":"REWORK","_number":3,"created":"2021-05-11 15:15:26.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/3"}}},"commit":{"parents":[{"commit":"baeaaa1f51825f895bf0817b8c45b80378a7af7a","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baeaaa1f51825f895bf0817b8c45b80378a7af7a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-11 15:14:31.000000000","tz":0},"subject":"Validate cipher in encryption spec","message":"Validate cipher in encryption spec\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0334b47c4039692e51af3c3929dd75bc86ee6b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0334b47c4039692e51af3c3929dd75bc86ee6b53"}]},"branch":"refs/heads/master"},"1223e817457a3507e3929c9630627bed1385a63c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-05-11 15:48:45.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/4"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-11 15:44:43.000000000","tz":0},"subject":"Validate cipher in encryption spec","message":"Validate cipher in encryption spec\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1223e817457a3507e3929c9630627bed1385a63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1223e817457a3507e3929c9630627bed1385a63c"}]},"branch":"refs/heads/master"},"f7506cdf978eb006625a8a97bea3d152c5c18c33":{"kind":"REWORK","_number":5,"created":"2021-05-11 18:54:36.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/5"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-11 18:26:00.000000000","tz":-240},"subject":"Validate cipher in encryption spec","message":"Validate cipher in encryption spec\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7506cdf978eb006625a8a97bea3d152c5c18c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7506cdf978eb006625a8a97bea3d152c5c18c33"}]},"branch":"refs/heads/master"},"c2030cbc5bc33714730d3cee3bb795d0b752cc50":{"kind":"REWORK","_number":6,"created":"2021-05-12 14:41:17.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/6"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-12 14:24:55.000000000","tz":0},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2030cbc5bc33714730d3cee3bb795d0b752cc50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2030cbc5bc33714730d3cee3bb795d0b752cc50"}]},"branch":"refs/heads/master"},"4349506465f379af9e19b09eab7d95187a12540c":{"kind":"REWORK","_number":7,"created":"2021-06-07 14:02:28.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/7"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-07 13:55:13.000000000","tz":0},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4349506465f379af9e19b09eab7d95187a12540c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4349506465f379af9e19b09eab7d95187a12540c"}]},"branch":"refs/heads/master"},"713bbb5bd0788e9c579a5ce568e4dc2f271cf3ff":{"kind":"REWORK","_number":8,"created":"2021-06-07 16:43:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/8"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-07 16:42:53.000000000","tz":0},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/713bbb5bd0788e9c579a5ce568e4dc2f271cf3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/713bbb5bd0788e9c579a5ce568e4dc2f271cf3ff"}]},"branch":"refs/heads/master"},"99a493944cc3cbef94c546d61f695c81ad09b5ea":{"kind":"REWORK","_number":9,"created":"2021-06-09 14:24:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/9"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-09 14:09:39.000000000","tz":-240},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.  This change introduces a\nminimum key size of 128 bits for encrypted volume types.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99a493944cc3cbef94c546d61f695c81ad09b5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99a493944cc3cbef94c546d61f695c81ad09b5ea"}]},"branch":"refs/heads/master"},"912dc2306f55ccaa88a6b8e643713a342c967d87":{"kind":"REWORK","_number":10,"created":"2021-06-10 16:29:32.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/10"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-06-10 16:29:14.000000000","tz":-240},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis introduces a new API microversion, validation does not\noccur on previous microversions.  This change introduces a\nminimum key size of 128 bits for encrypted volume types.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/912dc2306f55ccaa88a6b8e643713a342c967d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/912dc2306f55ccaa88a6b8e643713a342c967d87"}]},"branch":"refs/heads/master"},"990ab3e3dad8ad82cd4aa113654335a56dc7f62a":{"kind":"REWORK","_number":11,"created":"2021-07-08 12:22:10.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/11"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-08 12:15:52.000000000","tz":-240},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis change introduces a minimum key size of 128 bits for\nencrypted volume types.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/990ab3e3dad8ad82cd4aa113654335a56dc7f62a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/990ab3e3dad8ad82cd4aa113654335a56dc7f62a"}]},"branch":"refs/heads/master"},"5d7142711b2301e3aecba3205610062dc124d311":{"kind":"REWORK","_number":12,"created":"2021-07-15 15:01:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/789603/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/789603/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/789603/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/789603/12"}}},"commit":{"parents":[{"commit":"0a1cc1a2d42d17ee5fcf3d273d543b598b033318","subject":"Abort volume creation when encryption spec is invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1cc1a2d42d17ee5fcf3d273d543b598b033318"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:21:42.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-15 15:01:14.000000000","tz":-240},"subject":"Validate fields in encryption spec create/update","message":"Validate fields in encryption spec create/update\n\nWhen the cipher is None or an empty string (allowed by\nour standard CLI clients), reject the encryption\ntype creation immediately, since the encryption spec is\ninvalid and will not be able to be used to create volumes.\n\nSimilarly, reject encryption cipher fields that don\u0027t\ntake the form of \"algorithm-mode-ivgen\", i.e. \"aes\" instead\nof the correct \"aes-xts-plain64\".\n\nBefore this change, such invalid cipher fields in encryption\nspecs would be accepted by the API, but volume creation would\nfail later with various errors.\n\nThis change introduces a minimum key size of 128 bits for\nencrypted volume types.\n\nCloses-Bug: #1926630 (2/2)\nChange-Id: I2414e521eb0b4ce48b41dfe4d305e002e8de8f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d7142711b2301e3aecba3205610062dc124d311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d7142711b2301e3aecba3205610062dc124d311"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
