)]}'
{"id":"openstack%2Fcinder~792462","triplet_id":"openstack%2Fcinder~master~Ibab932d3536d44dd7414560768ed5bdeaa9e61e1","project":"openstack/cinder","branch":"master","topic":"bug/1929219","hashtags":[],"change_id":"Ibab932d3536d44dd7414560768ed5bdeaa9e61e1","subject":"[Pure Storage] Fix minimum SDK version required","status":"MERGED","created":"2021-05-20 17:37:41.000000000","updated":"2021-06-30 18:04:21.000000000","submitted":"2021-06-30 18:02:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"792462-bug/1929219","meta_rev_id":"113c17f77394fd3919712b996bfd31d18e260cca","_number":792462,"virtual_id_number":792462,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 18:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-29 15:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2021-06-30 16:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2021-06-30 16:49:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 17:56:28.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 18:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 18:10:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 18:18:18.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-05-20 18:32:54.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 18:59:05.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-20 22:35:45.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 22:41:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-20 23:05:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-20 23:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 01:33:06.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 09:43:16.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 10:43:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-21 17:10:31.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-21 17:43:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 22:17:33.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-05-21 23:00:17.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-05-22 00:04:51.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-05-22 02:00:26.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-05-22 05:37:44.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-05-22 08:07:07.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-23 07:04:31.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-24 20:04:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 00:51:11.000000000","updated_by":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"reviewer":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"state":"REVIEWER"},{"updated":"2021-05-25 11:29:33.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 19:18:46.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-06-18 14:29:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-18 15:30:08.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2021-06-18 20:59:01.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-21 05:20:24.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-06-21 15:50:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-21 18:10:39.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-06-23 14:00:03.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-06-28 21:12:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4c798ecc0a86f1fce3db760c4cc5ba35a3e74156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-20 17:37:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89990243a9b95a65e73824226505fb67d9be891c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-20 17:56:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/1/check/cinder-storpool-tempest/db3e1db/ : FAILURE in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2e5843b57e04fdc29cabca2a6af3ce01c71d7584","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:02:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffb2acaf3e174374b516780f0a5b161a : FAILURE in 20m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d23efa44e384a4fbdb10a5f9b81cb15 : FAILURE in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ba8debbc35957dc75820d43fa40befff09c0e8a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 18:10:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/1/check/iscsi-driver-primera/af086df/af086df.html : FAILURE in 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7ea1c67b30246f257c2c4b4319ea50378c0de4ac","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-20 18:18:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f8ca482ae1bbb8769c07c0b9aaa3663e0bc07b3e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-20 18:32:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/792462/1/ : FAILURE in 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"178c7f33346f4d31a0b6bdb735c870b209a48271","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 18:40:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/792462/1/check/iscsi-driver-nimble/044f788/044f788.html : FAILURE in 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"36812185e3e9e63f275a1d0d9caa840e728c9314","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-20 18:59:05.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/1/iscsi/ : FAILURE in 1h 20m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/1/smb/ : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4a7235d60cc1f46a96d4ffea9642df3f70d9e195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-20 19:01:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55b21d11bcef769219740b1089bf134dfea04513","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-20 19:10:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/2/check/cinder-storpool-tempest/405f00c/ : FAILURE in 7m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a19f1f5f9767ec91a028f5e04e49d89f145119b8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-20 19:29:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/792462/2/ : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cf678ac658d0d8d16fbd06262e1219107139213d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 19:35:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bb30541d5e24dbe8cefca1311b50343 : FAILURE in 29m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3af0e3977df14a1f814b3fc48480f496 : FAILURE in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11b425b88366a87e7183fef51c4abcd814f7f63b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 19:41:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/792462/2/check/iscsi-driver-nimble/5bbf167/5bbf167.html : FAILURE in 28m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bb42b265a53dedb25370f9be949753144e5674dc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-20 19:58:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/2/iscsi/ : SUCCESS in 52m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/2/smb/ : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":2},{"id":"050d69ea1441b2bc0fe0d637d3cb274c46cbbcf3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 20:02:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/2/check/iscsi-driver-primera/eddae99/eddae99.html : FAILURE in 50m 21s","accounts_in_message":[],"_revision_number":2},{"id":"7128b2262b5eba198466978cbea914b09091bd63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-20 20:22:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d7ffacf08b130a2548183b45534f8b14b6d4d68","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-20 20:23:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f1cf086dd4735f0007bb28d6aa34ae9aab1b6d92","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-20 20:31:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/3/check/cinder-storpool-tempest/7be3c81/ : FAILURE in 7m 32s","accounts_in_message":[],"_revision_number":3},{"id":"c2846368d75a08c9d32ae0f85b3572ea5c932499","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:44:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cf5dd9a33a1452f97e5f2c4c0c0ca73 : FAILURE in 19m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90af137e3a624fcea5bd1ab94585c0c4 : FAILURE in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de6d8e1c24c72af95562e0382d3b11eb88f67c86","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-20 20:51:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/792462/3/ : FAILURE in 27m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a81581c6eacf9bf2f6374f6bd4ffa08d3a8313e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 20:55:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/792462/3/check/iscsi-driver-nimble/5ab18f2/5ab18f2.html : FAILURE in 31m 33s","accounts_in_message":[],"_revision_number":3},{"id":"2a2ffa1cd7ffd947a1ffbd521a737ecd1b231c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-20 21:02:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad8001453b55167e67f21a29d2874e35d5146d0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-20 21:11:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/4/check/cinder-storpool-tempest/8f2f874/ : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f7806668cf18ce65643ed10c06d5839308933bf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 21:21:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29e36bdc01af4d708a13cd24a0db9a55 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53833d9c7ad54e9ba4261a5ae51868ee : FAILURE in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51a1c71d3de3decf4dc9a2f8748f00dc26e1c690","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-20 21:29:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/792462/4/ : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":4},{"id":"e86880f134ec04db81946cf7a996399138386171","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 21:30:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/792462/4/check/iscsi-driver-nimble/753ea59/753ea59.html : FAILURE in 26m 51s","accounts_in_message":[],"_revision_number":4},{"id":"25f00bcd1f9bcfdba3e8fb9df57b5928863cf6ef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 21:55:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9c011d9/9c011d9.html : FAILURE in 47m 57s","accounts_in_message":[],"_revision_number":4},{"id":"e08ea850340ccd3c1bc454967214428678a83f74","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-20 21:57:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/4/iscsi/ : SUCCESS in 54m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/4/smb/ : SUCCESS in 52m 16s","accounts_in_message":[],"_revision_number":4},{"id":"8e1b93b1a867d7b800448013859112d488a43bcf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-20 22:04:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/4/check/iscsi-driver-primera/6902e54/6902e54.html : FAILURE in 55m 20s","accounts_in_message":[],"_revision_number":4},{"id":"4417ecc41df1c7ca3052f372a7cf4a4557a533c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-20 22:35:45.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/91f5d6c9eeb34fc99bba8fb8954de078 : POST_FAILURE in 1h 20m 55s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7adc36bef2534dc088b4fa9eec2027f7 : POST_FAILURE in 1h 24m 43s","accounts_in_message":[],"_revision_number":4},{"id":"90b0c9f3e18f0cbdeda8ee990e18c6ca084d058d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-20 22:41:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9273079/ : SUCCESS in 1h 37m 42s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6e59c0f/ : SUCCESS in 1h 26m 02s","accounts_in_message":[],"_revision_number":4},{"id":"98f0190705f0a29532b9869062feaeefc3bb435f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-20 23:05:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/62/792462/4/check/dsvm-tempest-cisco-zonemanager-job/0197bd8/ : FAILURE in 1h 17m 00s","accounts_in_message":[],"_revision_number":4},{"id":"39a588bffd695cf5eda0109c3a362848e851c4c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 23:22:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d93ff921ffa42fbb6793e2f0ec17816 : FAILURE in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df29401dd34e4660a75c300f014d11f4 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26e8b710c7aa4f23a11ca9acf9c903c6 : FAILURE in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e67a9fd8f0434df58fdfe1958af0d13a : FAILURE in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f608a1f33094bdfa9c804f44f161543 : FAILURE in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cbce326a2844a78f8b5f620c23d1cd : SUCCESS in 11m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e06f57b7de0244f1a83e1a5b513f472f : SUCCESS in 3m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/ae989540ddc346f59dc6197dbbe8690a : SUCCESS in 1h 06m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4b0b10dd4684277a36eda2eb8417582 : SUCCESS in 1h 08m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/747ad695deb1420f8366a0289c537e25 : FAILURE in 12m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b07038303c5411a8111dca924465cdb : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/273d5459753e4e198ae6016628a6834c : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4742b941487f40c89da5d02e4a8a1c15 : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0f59b9dfa004fb0b5546df27ed781d9 : SUCCESS in 39m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ffc2a8b51dc84b25830b3ea3896d4d4d : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7f096d67ffab4cc3a5c9452932730095 : SUCCESS in 1h 07m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8968f9b773ad4660bbd21c50d340229f : SUCCESS in 1h 23m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c60b8eba59a645e1951f05f83158e65b : SUCCESS in 1h 36m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fe88d6f6611d421ca1d15b3a23f23ce9 : SUCCESS in 1h 11m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a25b112aa1bf4255aca78c6882c4c8f5 : SUCCESS in 1h 11m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57c18c9466334a2fa245bfc7c20ed0d1 : SUCCESS in 1h 13m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/663f70d8c46e4774b1db944a4538b879 : SUCCESS in 2h 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37f2b5e637ed4b6ca25d633d1aa01a54 : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":4},{"id":"35b8774541e0562bba00d3756666f370089063d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 01:33:06.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-1/ns5_nfs : SUCCESS in 4h 48m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-1/ns5_iscsi : SUCCESS in 4h 54m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a46414d718c34fbd957691dc221a5b1dc0ccf1f8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 02:56:17.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-2/ns5_nfs : SUCCESS in 4h 42m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-2/ns5_iscsi : SUCCESS in 4h 48m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f51c1616f6d1772c974344e361d22eea62ab11ca","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 06:34:12.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-3/ns5_nfs : SUCCESS in 4h 36m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-3/ns5_iscsi : SUCCESS in 4h 40m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8cb1ea5e24248b87e6726e158198747f25096a6e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 07:15:31.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-4/ns5_nfs : SUCCESS in 4h 31m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-4/ns5_iscsi : SUCCESS in 4h 37m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"fef541cc40064cffbc8228503a63821702f4dd83","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 09:43:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/62/792462/4/check/DellEMC_XtremIO_FC/4b840d5/DellEMC_XtremIO_FC/187 : FAILURE in 1h 04m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/62/792462/4/check/DellEMC_XtremIO_iSCSI/7ed162e/DellEMC_XtremIO_iSCSI/194 : SUCCESS in 1h 48m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"995c10b3318b57db2cb4af6a5d83df677b0521d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-21 10:43:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/62/792462/4/check/DellEMC_Unity_FC/a50144b/DellEMC_Unity_FC/200 : FAILURE in 1h 33m 07s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/62/792462/4/check/DellEMC_Unity_iSCSI/2c00339/DellEMC_Unity_iSCSI/186 : SUCCESS in 1h 48m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"576bca9b92d7acb21acd23e6f82eed4639414d0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-21 13:51:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5565a86c0ff2bdb0ecc606d9cbae936e13bf2b18","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-21 14:00:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/5/check/cinder-storpool-tempest/7586e53/ : FAILURE in 7m 32s","accounts_in_message":[],"_revision_number":5},{"id":"653facf8b141e9969b0d8d76800d7c65babdec97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-21 14:12:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9f01522301b61a935d50c6bcb172500271b26ac3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-21 14:22:03.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/6/check/cinder-storpool-tempest/0d61748/ : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":6},{"id":"7256b39ab125c1d2f0307a77e10ee3abf46525eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-21 14:36:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"94f1838e7f96d122225020805b48802b511261f0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-21 14:46:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/7/check/cinder-storpool-tempest/ed15e9c/ : FAILURE in 7m 34s","accounts_in_message":[],"_revision_number":7},{"id":"0d724078b1909214ed68b70134680b2f54f40055","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:00:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b695dc622bd4fb99557779ff983f486 : FAILURE in 20m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3081135b2c34011820b6f84a0189f66 : FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f827e11526b77017bf37aa42582239f6f7d005fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-05-21 15:01:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5d856350f10665d9057054956b2a0490cb008b27","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-21 15:17:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/62/792462/8/check/cinder-storpool-tempest/0f04cb1/ : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":8},{"id":"fae5c44fe07833290b08487fc4509df5d4506b62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:26:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c9e19a4ad79498ea8eef643ce13b256 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0c53c5c02d44e4a9f2b91b631d753f5 : SUCCESS in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a80ce49154b027e04c2086ace25db0720f77ff39","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-21 16:29:04.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/8/iscsi/ : FAILURE in 1h 07m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/8/smb/ : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":8},{"id":"16191cdd88be655fab6d8a7f0dba448d6d8c2fce","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-21 16:44:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/792462/8/ : FAILURE in 27m 33s","accounts_in_message":[],"_revision_number":8},{"id":"d34f635c156a6bceb832ace1c16080913cea47eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 16:57:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fbd60fbeb0641e2810476455efd83df : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36ed5e2e6b642d2984f33537de09f6c : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ced0fadfb7c4889b15253650b80829d : SUCCESS in 12m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82eb1b88acf043649df819dc04c49891 : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbbd830d09744650a1c2795c1eb9d36a : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37602fb97dfb47c980815b5699222f7e : SUCCESS in 13m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5e20cd5561e4d6b90e224e7ef33ce08 : SUCCESS in 4m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/6929abc9db224a148856ff77dce17a54 : SUCCESS in 1h 02m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba5c26a712684261a6362ebfe434b648 : SUCCESS in 57m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bbe9c9187948454fbf9aa53efb50bf8c : FAILURE in 12m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e23ec34190849639368756b0ea2b96d : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbaa1ba81f61483881f63d59e8d1a54b : SUCCESS in 7m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2265f65b6252412483c53ccfb907f4c7 : SUCCESS in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05d88590b9e240559d792b2c7a3f90a7 : FAILURE in 28m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2bbddd7b73b24e03bfcc31527102e5e4 : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2bb2e157e82445989c97a213b70ac6df : SUCCESS in 1h 07m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/54ed1d365faa43d094d9f75f08beb6ca : SUCCESS in 1h 17m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb329993daa9481583101fd481eec8b1 : SUCCESS in 1h 25m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60a80c86c4b841b1b5b4c7fba52cd52c : SUCCESS in 1h 11m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d252dac7517344d7b59cd2da1cb46a40 : SUCCESS in 1h 23m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/09e568e628b14900a5203d788cd0fbcc : SUCCESS in 1h 18m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40726db573d1497b94b8386650a95a69 : SUCCESS in 1h 55m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/983b26d264524133b841371da6926b29 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":8},{"id":"662e4702c3a0e3cbfce23d82a672b07a38434b4f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 16:58:42.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/8/check/iscsi-driver-primera/e7d1b15/e7d1b15.html : FAILURE in 34m 21s","accounts_in_message":[],"_revision_number":8},{"id":"b4b0e48e6bd3a9fb7f637372c1be892db6a55a41","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-21 17:10:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/62/792462/8/upstream-solidfire-check/cinder-solidfire-iSCSI/62023c5/ : FAILURE in 11m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"cfa35375323bcc4525575cf1ff4b15b0b6b42d46","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 17:29:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/792462/8/check/iscsi-driver-nimble/269865f/269865f.html : FAILURE in 27m 12s","accounts_in_message":[],"_revision_number":8},{"id":"5483f2fff94c5188cd61ccd313d0d91c20a1a958","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-21 17:32:10.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3e3db8a/ : SUCCESS in 2h 23m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/70c8466/ : SUCCESS in 2h 14m 21s","accounts_in_message":[],"_revision_number":8},{"id":"2fff2001506adf862f9b64c8178d1c2485954986","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-21 17:43:49.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/62/792462/8/upstream-check/cinder-cDOT-iSCSI/a33eb24/ : FAILURE in 10m 20s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/62/792462/8/upstream-check/cinder-cDOT-NFS/b45e6f8/ : FAILURE in 44m 54s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/62/792462/8/upstream-check/cinder-cDOT-FCP/25dd898/ : FAILURE in 18m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"827c19271f28306ca53cb496479c75dbcdac82bb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-21 18:10:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/62/792462/8/check/dsvm-tempest-cisco-zonemanager-job/89f8ec2/ : FAILURE in 32m 13s","accounts_in_message":[],"_revision_number":8},{"id":"cb15f9562e96bad585450163cf01722c8cf1b058","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 18:28:28.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-5/ns5_nfs : SUCCESS in 4h 31m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-5/ns5_iscsi : SUCCESS in 4h 36m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"4f894af9ea731d26b6ce76f8aab7d3c80a731adb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 18:43:37.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/7ba3ba9/7ba3ba9.html : FAILURE in 53m 47s","accounts_in_message":[],"_revision_number":8},{"id":"cf471c2f4bf045eb044e71ec60dcf0640ca9f671","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-21 18:54:35.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/13d05be4338947c399b291d06809becb : POST_FAILURE in 1h 46m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/609d6e6e7e5648f8a72c0b1deabaa3f3 : POST_FAILURE in 1h 46m 37s","accounts_in_message":[],"_revision_number":8},{"id":"0c90bcd727b8682e1e692b1894b8847432a94a93","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 19:20:52.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-6/ns5_nfs : SUCCESS in 4h 37m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-6/ns5_iscsi : SUCCESS in 4h 44m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"dda07a59c379828c4840701ee57fe5dca2af3cb6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 21:00:34.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-7/ns5_nfs : SUCCESS in 4h 50m 33s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-7/ns5_iscsi : SUCCESS in 4h 53m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"4c68bade7c9e0a6ca3cc2cb742d66dbe891dd7a7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-21 22:17:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/62/792462/8/check/DellEMC_PowerStore_iSCSI/50e4bea/DellEMC_PowerStore_iSCSI/117 : FAILURE in 47m 23s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/62/792462/8/check/DellEMC_PowerStore_FC/772572e/DellEMC_PowerStore_FC/118 : SUCCESS in 1h 00m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"7def1fc99757a923f23a2e0e1ca5e0357db45a75","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-21 23:00:17.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792462-8/Seagate-iSCSI : SUCCESS in 43:01.81\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792462-8/Seagate-FC : SUCCESS in 39:18.97","accounts_in_message":[],"_revision_number":8},{"id":"d2363ece6d8faf6763fa805f6b99ae708e96835c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 23:27:21.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/62/792462/8/check/DellEMC_XtremIO_FC/7b67cfb/DellEMC_XtremIO_FC/195 : SUCCESS in 1h 24m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/62/792462/8/check/DellEMC_XtremIO_iSCSI/6b7ef61/DellEMC_XtremIO_iSCSI/200 : SUCCESS in 1h 57m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"662b580807af1cec45368758c3adbd0370921c95","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-22 00:02:41.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/62/792462/8/check/DellEMC_Unity_FC/345fff7/DellEMC_Unity_FC/206 : SUCCESS in 1h 28m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/62/792462/8/check/DellEMC_Unity_iSCSI/3a2eafa/DellEMC_Unity_iSCSI/192 : SUCCESS in 1h 46m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"e04ae9eb68780a218a2861aeae7744da261b805a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-22 00:04:47.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-8/ns5_nfs : SUCCESS in 4h 42m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-8/ns5_iscsi : SUCCESS in 4h 46m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"5f8c5e03d4be615cc99fb550efdbda0e42ec66b2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-22 00:04:51.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792462-8/Lenovo-iSCSI : SUCCESS in 52:09.67\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792462-8/Lenovo-FC : SUCCESS in 46:31.52","accounts_in_message":[],"_revision_number":8},{"id":"d57dc19964cf70cfa5e12e6789c139f81ac8f3d0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-22 02:00:26.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792462-8/HPMSA-iSCSI : FAILURE in 1:26:30\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792462-8/HPMSA-FC : SUCCESS in 33:52.15","accounts_in_message":[],"_revision_number":8},{"id":"643cd444f21067a13b08a86c17086744d478c940","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-22 05:37:44.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/62/792462/8/check/DellEMC_SC_FC/9fa5ba8/DellEMC_SC_FC/133 : FAILURE in 1h 12m 38s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/62/792462/8/check/DellEMC_SC_iSCSI/3a999d1/DellEMC_SC_iSCSI/130 : SUCCESS in 2h 15m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"25eaee322519067f2fc0e4453274a12496c83203","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-22 08:07:07.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/62/792462/8/check/DellEMC_VNX_iSCSI/f4143ad/DellEMC_VNX_iSCSI/134 : FAILURE in 1h 12m 42s\n- DellEMC_VNX_FC http://publiclogs.emc.com/62/792462/8/check/DellEMC_VNX_FC/b29cc8f/DellEMC_VNX_FC/139 : SUCCESS in 1h 56m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"9e165007d00af637019766274af44b5a10e23fa7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-23 07:04:31.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/62/792462/8/check/DellEMC_PowerMAX_FC/4a3db57/DellEMC_PowerMAX_FC/94 : FAILURE in 4h 27m 47s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/62/792462/8/check/DellEMC_PowerMAX_iSCSI/5d9cbde/DellEMC_PowerMAX_iSCSI/84 : FAILURE in 5h 16m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"14bddc46d38875d4bbb038219aa275ac1c2a4adf","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-24 09:09:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/792462/8/ : SUCCESS in 1h 53m 00s","accounts_in_message":[],"_revision_number":8},{"id":"74978da3a9818c8b53a20d23faecbee41967a374","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-24 20:04:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/8/20210525022319/NEC-ISCSI : SUCCESS in 2h 35m 30s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/8/20210525022319/NEC-FC : SUCCESS in 2h 40m 42s","accounts_in_message":[],"_revision_number":8},{"id":"1802e393d58801c661b0837902aceb1812d95f65","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-25 00:51:11.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"a8f85f57c85e74b4f9a7f1014f7da24c8a886918","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-25 11:29:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/62/792462/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0242865a04201af121a212b45678ddd173d4ad3d","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-11 19:57:56.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"72686f0427c5940b37c04a5176066fe46b0ec2cb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-11 20:48:48.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"45f54de3c973a03dff33d91e3e0dc3fb30b315b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-11 20:49:20.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"97e0b7f197c39d6a5c917233127033a82f9bdf40","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-11 20:52:27.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"fd9e282933b198c95ad1cf51bc30fa5d822e12b5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-11 20:52:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"33b46e56df0c82e0209fa1e8902121bec40dd083","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-14 17:34:49.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"4620b9083f619be5fa43e9308bc16c6f21c638d2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-14 18:19:43.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/7789b502361e4353903a4123b3e193f4 : POST_FAILURE in 32m 46s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/cd9348cbdf104974b3523304d26926f1 : POST_FAILURE in 43m 53s","accounts_in_message":[],"_revision_number":8},{"id":"1861fced1d2bcf94c0a8cc1319b2e3f8a546b51e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-14 18:27:10.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"2302fec037e8e36484634c3af04944781efa918b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-14 20:30:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/5be079aa7e214aa0b7e6aba40af3bae5 : POST_FAILURE in 1h 34m 46s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/80ad2cee391f403ba036cce90deadc6c : POST_FAILURE in 2h 02m 11s","accounts_in_message":[],"_revision_number":8},{"id":"e49bd3e0347a8f9a5a4af2f7b2463efc8060cf09","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-15 15:59:46.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"5c956611851e43b6260b9b81906b2ab5ae02e19b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-15 17:13:53.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"f6061effe3093a507bb5bbcec1fb1492df60becc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-15 18:59:56.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/1256741331d14594b88a7450e532c1d4 : RETRY_LIMIT in 17m 44s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/5eee74513d684e26859678db1ab6d33d : RETRY_LIMIT in 43m 56s","accounts_in_message":[],"_revision_number":8},{"id":"b8dfd56c5f93c0b8c70de574dd574f7a752602e0","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-16 14:44:02.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"d41dd64d6d3867a989f2dcf4c0f7c60b9a28a31f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-16 15:20:17.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"d3c9e1017ad487896db6b8299691d6b6c0c0ccc5","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-16 15:25:09.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"93c1f0dcc680eab8c659bce620849a92e6c100d9","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-16 15:54:33.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"e4a7acdadaba59da789c2170878fb723acc3f84d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-16 15:58:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1a3674b/ : FAILURE in 16m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/51c1e8a/ : FAILURE in 17m 22s","accounts_in_message":[],"_revision_number":8},{"id":"e7ca869029a0e49be59d63f3a74bf4b4d58efda3","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-16 16:11:34.000000000","message":"Patch Set 8:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"58b11638bbf91d729cb72e695f2615bc4bf66775","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-16 16:46:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2de84e6/ : SUCCESS in 33m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ea77aeb/ : SUCCESS in 30m 26s","accounts_in_message":[],"_revision_number":8},{"id":"e48de0dd206843505f038858719c105ff29135ce","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-17 19:18:46.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nOk, a couple of things.  Two things in-line that I commented on.  Also, could we add a comment to the commit as to why all the formatting changes.  Also, I think this deserves a release note.","accounts_in_message":[],"_revision_number":8},{"id":"dd53e8a3a801dbaf44100331b496ef21d84e5354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-17 19:24:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c4e3caf6a721ab903fbd3f7b08260710e4615bb8","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-17 19:25:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"56b9f21f6135a84aba7e544f536777c3b44e27fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:48:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3612152b24a4d9691ab5ddac005a66a : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/854e84fde2104c3fa9ac2f6b7e38d1fd : SUCCESS in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4a9332017965c2c7bcef7afad74ecc2e67b1e1e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-17 19:53:08.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2eeae3a/2eeae3a.html : FAILURE in 26m 56s","accounts_in_message":[],"_revision_number":9},{"id":"e86b9ee5a461fc1cd4a3bf5b2ef847a9d31778f1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-17 20:25:49.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/9/check/iscsi-driver-primera/5806889/5806889.html : FAILURE in 59m 22s","accounts_in_message":[],"_revision_number":9},{"id":"2dfb3aadb9122ebf245af20b1a5abec7ee318a97","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-17 20:29:32.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/62/792462/9/check/dsvm-tempest-cisco-zonemanager-job/8b08565/ : FAILURE in 59m 34s","accounts_in_message":[],"_revision_number":9},{"id":"32d05f6ac229ecbbac095381a82cb808915d981b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-17 20:30:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/9/iscsi/ : SUCCESS in 55m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/9/smb/ : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":9},{"id":"c1487a1a490198d231bf760c25a4cffd4a538880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 21:27:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a43abcbc51c34176a44ee6ca82aa6889 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b231d2bafa7e4f4d8f8408f66c5856cc : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eda4ebafdb1407b907e357c390f436e : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f57bec020bb4d48b2e74d86a86fc72a : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd09f9733234aa6990f026703c40454 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c4c9db529049a99504eefa414d8338 : SUCCESS in 13m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/94ea13b77634425da45eae0d8b2190a9 : SUCCESS in 4m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/b743ef8cbf6342e9900a65a7557f4227 : SUCCESS in 1h 16m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a6f875a9be24cd08d81632cffaf3a49 : SUCCESS in 55m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/851c84729ae440d2b3d27eadecf50e6e : SUCCESS in 11m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/682a9da3fa304a6bad8e8a1fdd9f352a : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86f7147a887f429da992d03e6648fd64 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e6fe8866988c42a5bc800e76e2ef71d0 : SUCCESS in 7m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0e92c1ae02904c1ab4105aaa7c2b03ac : FAILURE in 25m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51eda2ae1de34915a93be51041028cc3 : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/338a017a8dcc42ab92fe624ccb852ac9 : SUCCESS in 1h 21m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c04bd66d8c7a4d4f86b7c307f5e889ac : SUCCESS in 1h 24m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea7ecdd2abfd4685b17985b26d076111 : SUCCESS in 1h 22m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c747e8001e114b37b5ed45cc6f86ba0b : SUCCESS in 1h 12m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f427503c164945f3945f2eb6c12cdf4c : SUCCESS in 1h 14m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/922f25d2740842298cb611cc6b37ae38 : SUCCESS in 1h 04m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9ba7a67d0034806a0211e7ee524066a : SUCCESS in 1h 57m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/305bbc8fd69d4764ae1b034380b5ce6f : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":9},{"id":"b896cfc193cea0724272ca2313758c540ffd471a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-17 22:26:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/62/792462/9/upstream-check/cinder-cDOT-iSCSI/4888450/ : SUCCESS in 39m 53s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/62/792462/9/upstream-check/cinder-cDOT-NFS/cf8e834/ : SUCCESS in 38m 03s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/62/792462/9/upstream-check/cinder-cDOT-FCP/230896d/ : FAILURE in 56m 21s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"b0d4693d1b4f3b36e0c202c99268667bb8e35cbb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-17 22:30:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/36d6668/ : TIMED_OUT in 3h 02m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a1254fe/ : TIMED_OUT in 3h 02m 45s","accounts_in_message":[],"_revision_number":9},{"id":"c985944ad2d5049f43255a56007ec6ca651750b7","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-17 22:36:01.000000000","message":"Patch Set 9:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":9},{"id":"c699a684a3962decf0a57fdece57c8a1a9f7c6f0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-17 22:36:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/62/792462/9/upstream-solidfire-check/cinder-solidfire-iSCSI/41cefd7/ : SUCCESS in 1h 05m 54s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"7b94337b4eb5d8aa8c26a0e38f993ad354669a0a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-18 00:52:21.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2f2f943/ : FAILURE in 1h 51m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9d17b07/ : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":9},{"id":"2cf98f1ce1416cc21a987affdf803f09bc3bad19","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-18 01:24:05.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/49bc0bbc6afe437ea2cc1998c57172ba : POST_FAILURE in 1h 31m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2526a7ce03f847bf8a8dcd45e636ae6f : POST_FAILURE in 45m 43s","accounts_in_message":[],"_revision_number":9},{"id":"9f6e3261235005868bd4fc017258912116ba8638","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-18 03:34:22.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-9/ns5_nfs : FAILURE in 5h 27m 56s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-9/ns5_iscsi : SUCCESS in 5h 29m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"e387d9bc6941f60f0ddbfa49f77e35298ce60aed","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-18 11:53:27.000000000","message":"Patch Set 9:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":9},{"id":"9196c6ef986969d99eb5b82a7c183c095fdd3ee1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-18 13:57:24.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-9/ns5_nfs : SUCCESS in 5h 1m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-9/ns5_iscsi : SUCCESS in 4h 51m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"33f5a0de5e141752ccf446a315592e110c03f399","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-18 14:29:11.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nLooks fine if you get rid of the crap changes. No reason for most of these (and we do not, will not, and have no desire to use black).","accounts_in_message":[],"_revision_number":9},{"id":"3ed1c301e676f0adf75b77f777b5859961580226","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-18 14:58:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6d66418/ : FAILURE in 1h 13m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6edc455/ : TIMED_OUT in 3h 03m 34s","accounts_in_message":[],"_revision_number":9},{"id":"68656f115202e620c1ecd5e16361d61d4c6b3f6f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-06-18 15:30:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\nDue to the formatting changes it\u0027s hard to tell what has changed in the driver.\nAlso, there are a few commented out blocks that are left in on purpose?  If they aren\u0027t needed\nthen they should be removed.  I like the reformatting changes from black, but they should be in a separate patch.","accounts_in_message":[],"_revision_number":9},{"id":"e64efe3732546a8dcebcb6a627ae6e5c592853da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-18 15:42:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f35e4918113974c3d722af22ed3ea0a5bed64425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-18 15:43:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3cf57a307b05e31b2acf0c6f1195094264ff9db4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-18 15:47:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1bbc50b7f03a83c47a05ec736f15a3a9419585fe","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-18 15:50:45.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"721a40332ce461930a9fe62ebda824a288ba04cb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-18 15:51:35.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 8: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Ok, a couple of things.  Two things in-line that I commented on.  Also, could we add a comment to the commit as to why all the formatting changes.  Also, I think this deserves a release note.\n\nRelease Note added","accounts_in_message":[],"_revision_number":12},{"id":"cd249d61a844450b4cac84f1aebe339052790340","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 16:08:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7ababd8a0db4ec59da4ace93733e889 : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36fd9c65c67c4b779e3b7920685e4507 : SUCCESS in 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bd1e3e5adca8046b9cac23861278c5f1401c9788","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-18 16:42:08.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/12/iscsi/ : SUCCESS in 53m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/12/smb/ : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":12},{"id":"1f722b8e0d5d4c764b666d93950f8a8fda7e3c3d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-18 16:49:22.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/62/792462/12/check/dsvm-tempest-cisco-zonemanager-job/c0461b6/ : FAILURE in 56m 24s","accounts_in_message":[],"_revision_number":12},{"id":"7d7811f20438160255bd19b22df849229e92d761","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-18 16:49:32.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/12/check/iscsi-driver-primera/43a441e/43a441e.html : FAILURE in 1h 00m 28s","accounts_in_message":[],"_revision_number":12},{"id":"a6bc4eedd1f3a742b82fd1228e4afe239e68a653","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-18 18:02:10.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4997115/4997115.html : FAILURE in 2h 09m 45s","accounts_in_message":[],"_revision_number":12},{"id":"4a4ce1194373963dc25e932985b13c103b6e1637","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-18 18:08:11.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/123ff31/ : FAILURE in 2h 18m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6d88e39/ : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":12},{"id":"3a613ccdad6e585a48ac3c4dd4ca0cd1c863b3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:24:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82196b92a66a423bb55253df9a3bb59c : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bdfc443e564d5997fa4fa1dab55ffd : SUCCESS in 15m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd542ccaeef74753af593badd75c2afb : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/550598e432cd4104aa1513f64c5196d9 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70f7a27d0ac14c769d48416423af6a40 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4a82917fb24ff08251d72407fb2ca9 : SUCCESS in 13m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c19e0ba0d894e03864e7b6cbf42c9fe : SUCCESS in 5m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/1537c0a746344d45ab2cdc464863133f : SUCCESS in 1h 10m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d724c0adbd724349a7d2f828bf1822f8 : SUCCESS in 1h 16m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2a530182daf4459b06c8fc2e66d7bb4 : SUCCESS in 7m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/74230a9f2162442a818d6d696d3d3ad5 : SUCCESS in 13m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7341693daf434a95a0cee81532840fde : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/41baad5b9ee847a5b5d7c36dadd8c2c5 : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42421bea00ae40ceb2241caf114b611b : SUCCESS in 8m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce529d2014c344eaac5289ce2035c7b0 : FAILURE in 39m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b7748d97f9942a7b6cb92ef6dd5e64d : SUCCESS in 8m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8257b99d633d4581bd8187801cd952d9 : SUCCESS in 1h 07m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0159f43c13d643bb94c53af1ce046ffb : SUCCESS in 1h 30m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecd3454624324ae186424bf330a8ebff : SUCCESS in 2h 51m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2f474a5b75794cd9ab9ba180ad497ecb : SUCCESS in 1h 18m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3522916ff6994ded97bcc0039616ec9d : SUCCESS in 1h 09m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/16109628824841d8a7ccaac1be35fb7c : SUCCESS in 1h 21m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9aebfe575dcb4cd084134790cf75bca6 : SUCCESS in 2h 05m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b685418d93a4c778c5b320a3f4e3a3c : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":12},{"id":"fcf0b0318a64cc8a592fcfd4a8604e2d30e86b1d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-18 20:22:02.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/62/792462/12/check/DellEMC_PowerStore_iSCSI/390414b/DellEMC_PowerStore_iSCSI/171 : FAILURE in 56m 06s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/62/792462/12/check/DellEMC_PowerStore_FC/025a1b3/DellEMC_PowerStore_FC/172 : SUCCESS in 56m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"f71a84c11b76c21864131c0245b090cffc08aff1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-18 20:25:00.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/884ef9e2bd624c259346d4e187bb0c25 : POST_FAILURE in 1h 42m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7cfaebed9b5d40a983e7eadc54842a8f : POST_FAILURE in 40m 03s","accounts_in_message":[],"_revision_number":12},{"id":"af1402aff25346976e1dd0640a470fb13014bb4f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-18 20:35:32.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/62/792462/12/upstream-solidfire-check/cinder-solidfire-iSCSI/bf9daf1/ : SUCCESS in 1h 08m 04s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"64f9ceadafb5b06fbbb29c1a92f96aba3f80476b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-18 20:40:12.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/62/792462/12/upstream-check/cinder-cDOT-iSCSI/a04bd84/ : SUCCESS in 45m 06s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/62/792462/12/upstream-check/cinder-cDOT-NFS/b88e8ac/ : SUCCESS in 41m 35s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/62/792462/12/upstream-check/cinder-cDOT-FCP/d3839d4/ : SUCCESS in 1h 12m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"6930895ecfd17bf47bb1d1532fe7cfc1a0816255","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-18 20:59:01.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/62/792462/12/check/DellEMC_PowerFlex/d08a07e/DellEMC_PowerFlex/173 : SUCCESS in 1h 32m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"967ff9a0c3d81dd4fe4401ab2c2ae99db848f10f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-19 00:41:00.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-12/ns5_nfs : FAILURE in 5h 53m 6s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-12/ns5_iscsi : SUCCESS in 5h 36m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"96b01d19f12c5d929758f02bf2a9f98021db3981","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-19 00:58:13.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-11/ns5_nfs : SUCCESS in 5h 35m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-11/ns5_iscsi : SUCCESS in 5h 36m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"902efa2fddbc27496ea2f86c3825c3823e5bf2b3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-19 01:07:48.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-10/ns5_nfs : SUCCESS in 5h 42m 28s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-10/ns5_iscsi : SUCCESS in 5h 37m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"f9918e2b36721df63a81f0334a94d73819079a88","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-19 15:13:54.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/62/792462/12/check/DellEMC_XtremIO_FC/406425b/DellEMC_XtremIO_FC/338 : FAILURE in 1h 22m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/62/792462/12/check/DellEMC_XtremIO_iSCSI/f4a238e/DellEMC_XtremIO_iSCSI/340 : SUCCESS in 1h 56m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"160c58b114e9103b9c1b2a9605217b46bb8ce2f2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-19 17:25:25.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/62/792462/12/check/DellEMC_PowerMAX_FC/ca02404/DellEMC_PowerMAX_FC/157 : FAILURE in 4h 32m 43s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/62/792462/12/check/DellEMC_PowerMAX_iSCSI/187a7b6/DellEMC_PowerMAX_iSCSI/143 : SUCCESS in 4h 06m 26s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":12},{"id":"089c4142855560f36f48e3d1a8d4f977ea520dc9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-06-21 05:20:24.000000000","message":"Patch Set 12:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/792462/12/fc : FAILURE in 0h 10m 51s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/792462/12/iscsi : FAILURE in 0h 12m 21s","accounts_in_message":[],"_revision_number":12},{"id":"abd666db0ed638f81ced667600f18809b864dcfa","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-06-21 05:20:50.000000000","message":"Patch Set 9:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/792462/9/fc : FAILURE in 0h 09m 47s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/792462/9/iscsi : FAILURE in 0h 13m 25s","accounts_in_message":[],"_revision_number":9},{"id":"d81aa649b1466c7815c3dfa0b54ae974e857323b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-21 05:32:05.000000000","message":"Patch Set 9:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792462-9/HPMSA-iSCSI : FAILURE in 57:52.00\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792462-9/HPMSA-FC : SUCCESS in 30:33.39","accounts_in_message":[],"_revision_number":9},{"id":"9ddf947c775cdfc95d29686493f92860fd0fb417","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-21 05:38:37.000000000","message":"Patch Set 9:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792462-9/Lenovo-iSCSI : SUCCESS in 47:56.62\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792462-9/Lenovo-FC : SUCCESS in 43:15.51","accounts_in_message":[],"_revision_number":9},{"id":"076813043708fc925f786e37295a9ba376cb7c8a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-21 14:16:00.000000000","message":"Patch Set 12:\n\nrun Pure Storage CI","accounts_in_message":[],"_revision_number":12},{"id":"5094e16112d92b374fdd1fe26de1e5dd85545fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-21 15:17:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2d77982f4e6fe9be4512caea47f5d54b74973360","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-21 15:18:16.000000000","message":"Patch Set 13:\n\nrebased","accounts_in_message":[],"_revision_number":13},{"id":"4cf7762d8ddbd2934a64140d5c4431150ceb4acc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-21 15:27:17.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"a6e6b37770920105df699936eb964e6b448e2561","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-21 15:33:20.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792462-12/HPMSA-iSCSI : FAILURE in 58:12.45\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792462-12/HPMSA-FC : SUCCESS in 33:01.24","accounts_in_message":[],"_revision_number":12},{"id":"091cb82f055b4ddb23af244da110a5c847c6cef0","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-21 15:34:53.000000000","message":"Patch Set 12:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792462-12/Lenovo-iSCSI : SUCCESS in 48:39.62\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792462-12/Lenovo-FC : SUCCESS in 43:45.75","accounts_in_message":[],"_revision_number":12},{"id":"5172e903166f54aacf080b843f5a7d3daa8ad86d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 15:36:52.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/325f3fa1232d4a26b2d6bce3d2974bf1 : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/967a3e5797314f919e4d5b37d204c2fb : SUCCESS in 14m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ded8e27953f3097426d5f1fa67e1a6de50607470","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-21 15:50:32.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792462/792462/13/check-not-vote/ext-cinder-zuul/e22a11a : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":13},{"id":"6cb91a0f27479e333bb386cf01c7da5b6ba040b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-21 16:13:36.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ff137fcaaf505b98da2bf2cfd04e69ea968bf482","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 16:41:29.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/871a848c11d343818177cdf8e9f40e6f : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f88f6a1959a4a8891b94bbffcd5906c : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6ae1a738e0a3cbbfc1e61a71b69d60ec26f62614","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-21 16:41:52.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"27f56fd6833efe0e8a822ebcde2be3053c2622bd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-21 16:42:45.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"8d3455d409373b48d4357dc48bef7642c8e78f56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:12:35.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/104515c4546c4310aa469fcd03cc7eac : SUCCESS in 22m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/814a8cb359d94725bfea9a65aad61dfc : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"499924010e73f941820158d636213bec6500e74c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-21 17:20:14.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792462/792462/15/check-not-vote/ext-cinder-zuul/6085f4f : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":15},{"id":"4b15aa0ee1085c0787494ee9ec67d2c54335f776","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-21 17:35:24.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/15/iscsi/ : SUCCESS in 49m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/15/smb/ : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":15},{"id":"79fd5e4460f91a01cefe6b41f432e2fa149a778c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-21 17:55:06.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/15/check/iscsi-driver-primera/9eb4229/9eb4229.html : FAILURE in 1h 08m 05s","accounts_in_message":[],"_revision_number":15},{"id":"fbb63f2f352799f8843d1b37b0cc887bf3f096ed","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-21 18:02:32.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/662d8ab/662d8ab.html : FAILURE in 25m 09s","accounts_in_message":[],"_revision_number":15},{"id":"9e9dc7efb6b856b03a594c5ceace4a73dfd8f238","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-21 18:10:39.000000000","message":"Patch Set 9:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792462-9/PVME-iSCSI : SUCCESS in 55:06.85\n* PVME-FC http://os-logs.tristero.net/dell-ci/792462-9/PVME-FC : SUCCESS in 52:37.73","accounts_in_message":[],"_revision_number":9},{"id":"3df5028e385aa5358c2f166c3946d1967fd20e75","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-21 18:19:04.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/962cce3/ : SUCCESS in 1h 35m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7d61a9b/ : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":15},{"id":"591aadf3f6b8f23c74654489564fd4a555e616ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 18:54:32.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb75d5e1199f411ab381f1b1c6cfbefe : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c24bfc07654ff8ba4be224eba58df4 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c73448396854b4f8784076066ef3b99 : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ddbb6998dd145ce8fc5c6049e1aea4c : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc803c60b47247aa9f260e7ce267e343 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c404df01114e4295397ae67bb84800 : SUCCESS in 12m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afac73417a904576ae6898ab6dc02726 : SUCCESS in 3m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/6c63a44f0c07472a903d229ed9980618 : SUCCESS in 1h 04m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a00474c52084dcc870c034cf3695cf4 : SUCCESS in 1h 05m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee1c57469ebf4f8fa6aedda7d9dae4b5 : SUCCESS in 6m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c1a962a0e9ce43a1a906a1a6cad11e80 : SUCCESS in 13m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11b4441cea8e459c85519a026e9c3c69 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a0c0331184a4e1d9d921951a66b56a8 : SUCCESS in 10m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51c414d06beb478e9ef1f03169a96662 : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c59a1c640f68446aa607ebf293490064 : FAILURE in 35m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06e0f3d003164d36b2cb4f6c90883ef7 : SUCCESS in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/437627cef83141e5967b9d3daae94e77 : SUCCESS in 1h 29m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a0015bfbf9a4d139ba818441d25ded2 : SUCCESS in 1h 27m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d7a78f628374f6d90f563c4ce086687 : FAILURE in 1h 18m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/18d0a8705de3421daf21be7a7f68c3a4 : SUCCESS in 1h 12m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8430167ece274dbda557ed253e1170ee : FAILURE in 1h 12m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d6b2b6c6d0054fb5a9e80b304d91ed50 : SUCCESS in 1h 09m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee7b9e9186f247558624a7183c16a107 : SUCCESS in 2h 10m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e76e37485cf7453a8537540347133541 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":15},{"id":"0a45dbed7b7555e5a04b2ad2c98ad1298b59eba8","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-21 19:06:31.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"bc68ab7d8963b81f34a7ca761adce7209b0eb3b9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-21 19:07:36.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"b088e4ee7f300037d656fb004b9cc61c1de5b4fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 19:29:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaf938d4295d41d48c0448dbe9244e4a : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8a8b0800e8f4bcab2ceff45ae83337a : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d9ce7c684a7fbee95c42caf737607823a3c234c7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-21 19:40:59.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792462/792462/15/check-not-vote/ext-cinder-zuul/3fcfbaa : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":15},{"id":"8be467aa30759896a22e9ab7c544df2e8f42d879","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-21 20:17:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/15/iscsi/ : SUCCESS in 1h 10m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/15/smb/ : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":15},{"id":"f6a0d74a45a2d26ff6205ee0d0b6b368905c79ee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-21 20:53:47.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c72f5b0/ : SUCCESS in 1h 11m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6cfed4b/ : SUCCESS in 1h 45m 25s","accounts_in_message":[],"_revision_number":15},{"id":"d8dddb3c91c2b4f478afcb8495fd8608dc3c9761","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-21 21:10:10.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/15/check/tempest-dsvm-full-focal-py3/020ee50/ : SUCCESS in 1h 52m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/bb07242d08934774ae2545a1a5ec32c4 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":15},{"id":"183ae6fec506ff483ed9a691e24af6f67dfb648b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 21:25:01.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3bb2c76a21e4755b143c3ec26c9b4e4 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0ceeed564c4c098a95804a4ddcba24 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f836be912b1b474188f519d2dbf4dbc4 : SUCCESS in 12m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e4ba186e9144188817f35960a451ce7 : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e7373b78c4546069073f093c94b2f87 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a2b8c1b9094cb39f10c53de84713c5 : SUCCESS in 13m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f11022c858a45e7b9fdc526224a8ccc : SUCCESS in 4m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2a90eb35cad64abdacafcd938c99c3d6 : SUCCESS in 1h 05m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/859e8a8a93ec4598ac2691d545b95a7f : SUCCESS in 1h 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16d343929aa2423591aad5cb8fd47e6a : SUCCESS in 7m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ad2b43ba16a4aa1a515a39211a0b66b : SUCCESS in 15m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03776227f7c2450f930f20a3b3a37fce : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bab9b64014f487483cdd21519119caf : SUCCESS in 8m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/15eec271e1874831b1aa2eaa6271a0df : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/811d8df7e89e46b691e715082ef6ff1d : FAILURE in 38m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64e3f127c7c343ce9cf4d32fa9c65d8f : SUCCESS in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84c27af287784efbafed1e705984a50b : SUCCESS in 1h 14m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a0ea47616e34fa09c27ac6f86382e79 : SUCCESS in 1h 35m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/535db73d0f6d45fa8c4639e76ced9d18 : SUCCESS in 1h 40m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb11253b0f634dd9a047104624c9c471 : SUCCESS in 1h 10m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dab39f74d6e948ecaa5fde9ece62a44a : SUCCESS in 1h 19m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/83e6ae75dcc54a638c0922aed937ceed : SUCCESS in 1h 22m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b33025a41c54623bd9580bab5a12476 : SUCCESS in 2h 13m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/342bac81df5a46958d398c399f83c3c3 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":15},{"id":"a93c2e405576d396c75fa40519edd7bad25ba458","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-21 22:41:40.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/62/792462/15/upstream-check/cinder-cDOT-iSCSI/06ee131/ : FAILURE in 55m 11s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/62/792462/15/upstream-check/cinder-cDOT-NFS/4d94dd8/ : FAILURE in 51m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/62/792462/15/upstream-check/cinder-cDOT-FCP/5b78f91/ : SUCCESS in 1h 15m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"06379731ce00888679d721308c4d2ebddf4b0454","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-21 22:58:15.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/62/792462/15/check/DellEMC_PowerFlex/acb496a/DellEMC_PowerFlex/189 : SUCCESS in 1h 32m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"d8c82126e60912b5a3309106174b6c82f12903e7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-21 23:04:03.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/62/792462/15/upstream-solidfire-check/cinder-solidfire-iSCSI/f39893c/ : SUCCESS in 1h 37m 27s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"3676f41fe8a254a3001c6faf2bb6579816f2f04e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-06-21 23:13:20.000000000","message":"Patch Set 15:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792462-15/Seagate-iSCSI : SUCCESS in 42:11.10\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792462-15/Seagate-FC : SUCCESS in 35:01.01","accounts_in_message":[],"_revision_number":15},{"id":"181fc8beb25d0b0ff27de91fe33409b3ca7fdd04","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-22 00:29:35.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-14/ns5_nfs : SUCCESS in 4h 44m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-14/ns5_iscsi : SUCCESS in 4h 52m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"44cd4bb336bfbf6094f8f5b37694e1fad4a22779","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-22 00:29:54.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-13/ns5_nfs : SUCCESS in 4h 48m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-13/ns5_iscsi : SUCCESS in 4h 56m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"f9924ef7557f89004ee0ac4fb017e7c8cc34d16b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 00:54:17.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/62/792462/15/check/DellEMC_PowerMAX_FC/43d3679/DellEMC_PowerMAX_FC/174 : SUCCESS in 3h 41m 54s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/62/792462/15/check/DellEMC_PowerMAX_iSCSI/ea5f47b/DellEMC_PowerMAX_iSCSI/160 : FAILURE in 4h 58m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":15},{"id":"627903c96877755b499f4a66068bed394f1966c0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-22 01:07:38.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-15/ns5_nfs : SUCCESS in 4h 44m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-15/ns5_iscsi : SUCCESS in 4h 49m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"72dda906fc5e1b083c1520c2cd82f23b9e85e187","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 02:18:29.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/62/792462/15/check/DellEMC_VNX_iSCSI/0988152/DellEMC_VNX_iSCSI/168 : SUCCESS in 2h 48m 37s\n- DellEMC_VNX_FC http://publiclogs.emc.com/62/792462/15/check/DellEMC_VNX_FC/e93ba20/DellEMC_VNX_FC/173 : SUCCESS in 1h 31m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"ffc44f2b7e1e3619272626d8de28d551c7cf8c4d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 02:43:21.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/62/792462/15/check/DellEMC_XtremIO_FC/ad9707d/DellEMC_XtremIO_FC/360 : SUCCESS in 2h 12m 38s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/62/792462/15/check/DellEMC_XtremIO_iSCSI/486d47b/DellEMC_XtremIO_iSCSI/361 : SUCCESS in 1h 53m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"3b65f97f81ca2fc39476fd2164ebf32b5ddc7009","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-22 05:43:08.000000000","message":"Patch Set 12:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792462-12/PVME-iSCSI : SUCCESS in 53:52.00\n* PVME-FC http://os-logs.tristero.net/dell-ci/792462-12/PVME-FC : SUCCESS in 52:41.63","accounts_in_message":[],"_revision_number":12},{"id":"67962fad807a837e16c70b33289229a0fae5efb5","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-22 20:27:51.000000000","message":"Patch Set 15:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792462-15/Lenovo-iSCSI : SUCCESS in 50:00.01\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792462-15/Lenovo-FC : SUCCESS in 43:56.21","accounts_in_message":[],"_revision_number":15},{"id":"fe6e271c54e4f421b2ca04f25e6ee5886cd5a4bf","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-22 21:19:34.000000000","message":"Patch Set 15:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792462-15/HPMSA-iSCSI : FAILURE in 55:27.69\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792462-15/HPMSA-FC : SUCCESS in 30:54.97","accounts_in_message":[],"_revision_number":15},{"id":"5c4d01fc199dd0671569ff10973bf800d856975f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-23 10:18:41.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/15/20210623163752/NEC-FC : SUCCESS in 2h 35m 35s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/15/20210623163752/NEC-ISCSI : SUCCESS in 2h 40m 45s","accounts_in_message":[],"_revision_number":15},{"id":"7a4b1601817bddc9eea241a4ed233bd7af906b1a","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-23 14:38:41.000000000","message":"Patch Set 15:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792462-15/PVME-iSCSI : SUCCESS in 56:17.21\n* PVME-FC http://os-logs.tristero.net/dell-ci/792462-15/PVME-FC : SUCCESS in 53:07.77","accounts_in_message":[],"_revision_number":15},{"id":"3819fc00b3930622350b34e5c090e00f7b988feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-23 20:14:35.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b18cf4343130d77f392e182a9ae6a7a95799a4b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-23 20:17:59.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f5065850e2d572fd7720b07a3080c65ee1e52c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-23 20:34:15.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"73a3346ced7cdcb71a110fa3be1915d4863c3cce","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-23 20:35:58.000000000","message":"Patch Set 18:\n\nCode un-blackened... This is now just a regular patch","accounts_in_message":[],"_revision_number":18},{"id":"50077ccc382369ba6c5bcf25b095ce3536a43f0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 20:55:18.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35b1a0ee08ec4ae7ba7e5a4932ca72e2 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4bc1e6c7b68450ea8f8a5275c223493 : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"aa53ff0ba8eeca31abef7a95804de0da049e3776","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-23 21:36:01.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/18/iscsi/ : SUCCESS in 50m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/18/smb/ : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":18},{"id":"3fba47723a0f9e0085c0b6f809c5a2e78cf9fcf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 22:56:38.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab626f5b744a459391f014c01eceeb0c : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fccb435d265445ebaab51d0da11e6681 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea4caccf7ecd42b18616e25110897a18 : SUCCESS in 10m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43babd4406cc4fdbbd1b0d5eac2608f9 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/180328e399934a808118168c58c2326c : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f86357999564bf5833a1613ba241554 : SUCCESS in 16m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b722f9a84aff47b9825ea730f7ed2288 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/cac2ee532418450cae3f83730d21cfd1 : SUCCESS in 1h 12m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92228d3ff9d14244aaa10ec9208bd076 : FAILURE in 1h 42m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28c1cddee22e43998e8ebf792765f111 : SUCCESS in 10m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/621e0f3df02e4e68aee7971a302e2331 : SUCCESS in 23m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59791facedf143a6a1b5e94a9468b0c1 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1f887dc191e456290eec25ee72f369d : SUCCESS in 7m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd39626fcb8349be8bd79a12fcf7d169 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f52991f1927a42898214f2c0f82a09d2 : FAILURE in 35m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ebcf4dd17fd34be19b3abd2120f4878d : SUCCESS in 7m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b02a18ca1ab34496b0c3f7ec4999c51e : SUCCESS in 1h 44m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a2c319fe3210400287e7f2bbcea1ace8 : SUCCESS in 1h 36m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d3d09ee058594400b67ecef9f26c0d34 : SUCCESS in 1h 38m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f204805c272748adbdcc78312cf79bd0 : SUCCESS in 1h 18m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dabcc6511cb54408b9e64dc37e2963b3 : SUCCESS in 1h 14m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8596368958af4f0b80daa189a133273b : SUCCESS in 1h 26m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a80d731287114d27a698943b4326a209 : SUCCESS in 2h 16m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b01e566b71844a52b2eb4d9603443160 : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":18},{"id":"cf5aa6d0cd6b02de4cca0797ea88eb5405454c68","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-23 23:06:36.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/18/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/95e0705/ : SUCCESS in 1h 18m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/18/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/316e136/ : SUCCESS in 1h 17m 22s","accounts_in_message":[],"_revision_number":18},{"id":"ea977bf9c89eeaf21242123ff73e79f3650acc1e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-23 23:15:19.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/18/check/tempest-dsvm-full-focal-py3/b1befa8/ : SUCCESS in 1h 45m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/18/check/devstack-plugin-ceph-dsvm-focal-py3/75ba7ac/ : FAILURE in 48m 02s","accounts_in_message":[],"_revision_number":18},{"id":"cadec69de99a56a5b758e70fef4c5a5f6b9fc0b5","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-23 23:33:38.000000000","message":"Patch Set 18:\n\nrecheck tempest-integrated-storage","accounts_in_message":[],"_revision_number":18},{"id":"40ea0938f2aacd88b45f859e3762dc8e6ff90c7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:54:50.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1de24dc06c0b462985dc6b0419234fb5 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af6f1df826f04cf1aceaabeb6c128572 : SUCCESS in 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"bea0524df6287b8c570fda5719b22d1a7ba35e63","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-24 00:17:28.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/62/792462/18/check/dsvm-tempest-cisco-zonemanager-job/c64c325/ : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":18},{"id":"eb5070b7a192aa1518506e67704f39f0c33972b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-24 00:38:03.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/18/iscsi/ : SUCCESS in 1h 03m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/18/smb/ : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":18},{"id":"a5afae38b1f1c1c65212204ce8c7e2855c57bc4b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-24 00:40:59.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/18/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8397f88/8397f88.html : FAILURE in 2h 44m 41s","accounts_in_message":[],"_revision_number":18},{"id":"91769cebc84975ef80f7e8d0599f86c70437c7e9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-24 01:29:52.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/18/check/tempest-dsvm-full-focal-py3/81cecb4/ : SUCCESS in 1h 46m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/18/check/devstack-plugin-ceph-dsvm-focal-py3/d3dd70d/ : FAILURE in 46m 39s","accounts_in_message":[],"_revision_number":18},{"id":"d6f13568e655a1594607f6ce8c861c0247a4f6a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 02:31:38.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9535090149574327a3463cefb88d4a6f : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e915bb6f464160910aef2691fe8c31 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b279105c7a41ce9c4086fe94500ca5 : SUCCESS in 17m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33769156e68a4a329749245b1855ba3f : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb541c434d44a16a31b91a78f4332f4 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63084dc756d44de7a19be57c37dd8160 : SUCCESS in 11m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c968f001c80646b59d986c56be35c6b3 : SUCCESS in 4m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e15cab3b373343f4a613d03d2400f828 : SUCCESS in 1h 00m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db0b3a88a02542458025269e351f4a81 : SUCCESS in 1h 00m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/568ff1e6ecdd4554aba3009b0208714c : SUCCESS in 7m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9d698ff1f54046088b1f83a64aee18fc : SUCCESS in 12m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf25b1117b394dc6b18ebab43109cd40 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c25afc054a4d498a8dcd894750cd236f : SUCCESS in 8m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/db19b37bfe4e4e539ad46d7ff389e92f : SUCCESS in 8m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e04480b317b24b8ab6f5c2138bb0bd3b : FAILURE in 25m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8db22b2dc87d4f57b5ee6f962eed4aa0 : SUCCESS in 8m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16b384836fdc42bf8832f8bceee3cfd3 : SUCCESS in 1h 21m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2630214ce6a1414f825abaa91d482fc8 : SUCCESS in 1h 33m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9c513a28c967430cbe431bf905285cfe : SUCCESS in 1h 17m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bfcbffc57f3248b89e94b8382d215ebc : FAILURE in 1h 10m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9c6defdddc94882981ddbd6f49fb6f4 : SUCCESS in 1h 09m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cb961a8153b5484392f2d7d789ed963d : SUCCESS in 1h 09m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/145c204bdedb436380c72eb3a22c96a0 : FAILURE in 2h 49m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f83a97be298e423db39762d9e6a1115b : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":18},{"id":"3db57048628b9b47690d8e019db1808d5e1cce84","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-24 03:18:35.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"f419575a71a69a7bcdaeac17364b1a8770a4f20c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 03:40:20.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9358334057374344b77ed0e2e4123b35 : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ee61defb92b4e67af7717b867c8f33d : SUCCESS in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"259c5340e026250b749de4fa3f20197b909ebab0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-24 04:11:17.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-17/ns5_nfs : SUCCESS in 5h 7m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-17/ns5_iscsi : SUCCESS in 5h 15m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":17},{"id":"25899136286740b632eeca100687e53a2694b5e4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-24 04:13:52.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-16/ns5_nfs : SUCCESS in 5h 11m 42s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-16/ns5_iscsi : SUCCESS in 5h 19m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"4fc1f03d5c4d682f0a0356a18ec6ccea9d0f91e2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-24 04:23:38.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/62/792462/18/check/dsvm-tempest-cisco-zonemanager-job/f5729f5/ : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":18},{"id":"66fd318987282db8742354f1ca3d4c09383b8d33","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-24 04:32:07.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/18/iscsi/ : SUCCESS in 1h 12m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/18/smb/ : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":18},{"id":"4902073593c084250e2ba058502c0dfb36656fe3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-24 04:35:01.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/18/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cc540bc/ : SUCCESS in 1h 14m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/18/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/612ff10/ : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":18},{"id":"a74bee16eafa280847bc74899b694141acf26b5f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-24 04:54:03.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/18/check/tempest-dsvm-full-focal-py3/da0b0f2/ : SUCCESS in 1h 30m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/18/check/devstack-plugin-ceph-dsvm-focal-py3/8472138/ : FAILURE in 35m 19s","accounts_in_message":[],"_revision_number":18},{"id":"803df8ae36e76fa4b045aa9e8ca54c6a5f8ed478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 05:21:11.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7198badea92745aa90a6479c7009aed9 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a063fe8c8014df8bbf83b56d961b4cd : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b88e310d98b64d5b863d820b7b71cad9 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f228c7c7de4470883a279c73320e4b3 : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5500ef5f6c0493a805f6e1fa108379a : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f475b19d914a06954d21ec07200d07 : SUCCESS in 13m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f762ef53da064c4ba00daad6c886bb19 : SUCCESS in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/60ed29eea77548029617d6fb33662c22 : SUCCESS in 1h 06m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4144d81e7747418ab515bea8afd93a54 : SUCCESS in 1h 08m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dd88ca9da99453ca9cf02a5858e5f89 : SUCCESS in 7m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f50b28c9f3164bfda4a8326c8554034e : SUCCESS in 12m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6168a1406dc4d8393d3ce969a03f7a1 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39687f462523408e83a4c8e565a3c4c1 : SUCCESS in 7m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8e8e18890ca04aac810aef586cea0f04 : SUCCESS in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/17897217d1974c81b7e373ae63bb2eb7 : FAILURE in 30m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/192c57db985a433bb0d8f014e26a79c0 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/90c7e8d9bb55470284cdfa1b186bf5fa : SUCCESS in 1h 07m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14511e072aa34f4c8b04891685818006 : SUCCESS in 1h 28m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f599c001a6e41c7b4930eb63f78d60a : SUCCESS in 1h 28m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c4a6522c8a52444e87db0cdb551f314b : SUCCESS in 1h 11m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48a7d8cba300458c846447c4a7604a95 : FAILURE in 1h 29m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a15cdd1000bb47e4a2b21aa584d1be91 : SUCCESS in 1h 06m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cedd7118e47f48118175002ec1b8dccd : SUCCESS in 2h 00m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/964d15709ca84841ba2a1676861758e3 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":18},{"id":"bb697a456a1d8f85505313ee462a2390105eaee4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-24 06:25:55.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/62/792462/18/upstream-solidfire-check/cinder-solidfire-iSCSI/af02469/ : SUCCESS in 1h 02m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":18},{"id":"6dc3d8b1618f74f9437b3e225056e70d985c68ed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-24 07:20:11.000000000","message":"Patch Set 18:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-18/ns5_nfs : SUCCESS in 4h 40m 42s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-18/ns5_iscsi : SUCCESS in 4h 45m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":18},{"id":"bba87eb85b317faf2baf7166c8a80b96a2e29cce","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-24 08:12:23.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/62/792462/18/upstream-check/cinder-cDOT-iSCSI/98463e7/ : SUCCESS in 34m 41s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/62/792462/18/upstream-check/cinder-cDOT-NFS/9216f90/ : SUCCESS in 35m 26s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/62/792462/18/upstream-check/cinder-cDOT-FCP/5363060/ : FAILURE in 17m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"a7eeed66e47bbc96923b202eedf305a61355ed2e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-24 09:26:23.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/792462/18/check/iscsi-driver-nimble/3340ca2/3340ca2.html : FAILURE in 1h 47m 33s","accounts_in_message":[],"_revision_number":18},{"id":"e27b977bfff723da184b967938c01d15505dfb7c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-06-24 13:10:35.000000000","message":"Patch Set 18:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792462-18/Seagate-iSCSI : SUCCESS in 41:06.55\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792462-18/Seagate-FC : SUCCESS in 36:49.40","accounts_in_message":[],"_revision_number":18},{"id":"94e82b4b5e31041b9d77b0dea7e5897287eaf539","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 14:20:30.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/62/792462/18/check/DellEMC_PowerFlex/875cafe/DellEMC_PowerFlex/220 : FAILURE in 1h 30m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"2ec94014d4e23041606a69f34a8911ecb41cff43","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-24 14:24:04.000000000","message":"Patch Set 18: Code-Review+2\n\n(2 comments)\n\nLooks fine and CI is happy.","accounts_in_message":[],"_revision_number":18},{"id":"c57c2a39df2e4c6a6bf5ecba177f990a47fc0ce2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-24 15:46:39.000000000","message":"Patch Set 18:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792462/792462/18/check-not-vote/ext-cinder-zuul/c5fbaa5 : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":18},{"id":"fb80aa3bf8e574f57f33b4b40cba3cd8d33548c6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-24 16:01:26.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/62/792462/18/check/DellEMC_Unity_FC/7be5141/DellEMC_Unity_FC/347 : SUCCESS in 1h 41m 22s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/62/792462/18/check/DellEMC_Unity_iSCSI/ec94efc/DellEMC_Unity_iSCSI/327 : FAILURE in 1h 58m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"07554222c74cd1c68cf639640bcab79941788215","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-24 20:05:53.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/62/792462/18/check/DellEMC_XtremIO_FC/4766379/DellEMC_XtremIO_FC/396 : SUCCESS in 1h 22m 30s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/62/792462/18/check/DellEMC_XtremIO_iSCSI/41eb812/DellEMC_XtremIO_iSCSI/391 : SUCCESS in 2h 43m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":18},{"id":"0a95eb0d45271f0f82f3506364bc298c65083d32","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-24 23:25:38.000000000","message":"Patch Set 18:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792462-18/HPMSA-iSCSI : FAILURE in 59:47.27\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792462-18/HPMSA-FC : SUCCESS in 33:06.45","accounts_in_message":[],"_revision_number":18},{"id":"3f36fd7c37275974cac3f145ddbd6e451f425735","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-24 23:33:30.000000000","message":"Patch Set 18:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792462-18/Lenovo-iSCSI : SUCCESS in 46:20.28\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792462-18/Lenovo-FC : SUCCESS in 42:24.59","accounts_in_message":[],"_revision_number":18},{"id":"e230ce99d6ae3cef7468c878a8b8392199f41e42","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-25 19:17:58.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/18/20210625234457/NEC-FC : SUCCESS in 2h 36m 08s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/18/20210625234457/NEC-ISCSI : SUCCESS in 4h 32m 58s","accounts_in_message":[],"_revision_number":18},{"id":"f8f85f58447490f666d7344ab78b28b82f93c3df","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-26 00:31:04.000000000","message":"Patch Set 18:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792462-18/PVME-iSCSI : SUCCESS in 54:25.66\n* PVME-FC http://os-logs.tristero.net/dell-ci/792462-18/PVME-FC : SUCCESS in 51:46.63","accounts_in_message":[],"_revision_number":18},{"id":"969d7498e619412477cb6f5af6d313799266331e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-28 21:12:05.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)\n\nComment inline about the releasenote; also, you need to modify setup.cfg to reflect the change in driver-requirements.txt.","accounts_in_message":[],"_revision_number":18},{"id":"f0eb0599a1cb91321f487ca4fb2458f377f9b616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-28 21:42:49.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"b20de01c580ddbba24e4d088ffcf0f5948ef268e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-28 21:45:20.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"07335c1c27122b986163f8af8f78b12b73bf3350","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-28 21:45:42.000000000","message":"Patch Set 20:\n\n(2 comments)\n\n\u003e Patch Set 18: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Comment inline about the releasenote; also, you need to modify setup.cfg to reflect the change in driver-requirements.txt.","accounts_in_message":[],"_revision_number":20},{"id":"1a95a9d5f4bc56509708a93d2e75cdcbd5fc9e79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 22:12:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae20b7191c604424b696ae66640ff3b5 : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/356e9222e19548d981db86b68217160c : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"aaaefeedfbbf21ecaef2a6ce9ec318755f5de7e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-06-28 22:41:01.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"fe7f326e2118615ce8f77487bff007d8c294ddea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 23:04:42.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1255bf1bead84662826a5edc0ab8e94b : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ce66b3556d941ec8b60273a50d804b3 : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"00a03b61bc84ef7ac6b4248b0f2fe14808c00fc4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-28 23:36:26.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792462/21/iscsi/ : SUCCESS in 54m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/792462/21/smb/ : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":21},{"id":"e9fdd4160830ac319b745c8194e87c8c229905d7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-29 00:00:45.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/792462/21/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ad93e9c/ : SUCCESS in 1h 17m 59s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/792462/21/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/10b05ef/ : SUCCESS in 1h 09m 06s","accounts_in_message":[],"_revision_number":21},{"id":"cb1417852a79e61a3c62edb9b8f25861e5b49462","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-29 00:05:14.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/62/792462/21/check/iscsi-driver-primera/a3948c7/a3948c7.html : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":21},{"id":"83696cc81c27bf144d6291def7692f9822ef6832","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-29 00:24:41.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/21/check/tempest-dsvm-full-focal-py3/c51483c/ : SUCCESS in 1h 19m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/62/792462/21/check/devstack-plugin-ceph-dsvm-focal-py3/3767414/ : FAILURE in 39m 49s","accounts_in_message":[],"_revision_number":21},{"id":"5c2fd21ce89fea5603c4cbb3fe3a77aad26aebe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 00:43:47.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d3b96e925aa7457688493402076ede94 : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5771c189c474b6db6a86bd6dfc1ca5c : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b1476a611d4e049cc3a7d441fd169d : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e416e5a5818d4a7f9bed78f3a4bbb010 : SUCCESS in 13m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a706269e8f24c8bb629c5a84abe5b6e : SUCCESS in 13m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/819218de9e684940a1b76c9d808ba18f : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cedfad789f449929da7d271d42b8195 : SUCCESS in 12m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6b512c846ba47d6a595e81317cc9914 : SUCCESS in 6m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/647987e892ca4ee58e5be767d99512a9 : SUCCESS in 1h 01m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/728805f089e647b19c4467f921c22889 : SUCCESS in 1h 06m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a78d9810db14039a044343889fa8eb1 : SUCCESS in 11m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/89c368be95594c25ad401e2edd57d3cb : SUCCESS in 15m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01632e632060426687533184280493ef : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c3476ea80344ce395d0cb4a9e1e8a80 : SUCCESS in 12m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd5d41ab2d874a3aac914d32b47457b8 : SUCCESS in 8m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a2af94eba7d4bc097d9a90d2ecaf398 : FAILURE in 33m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52b69f0445044372ab947bfb087de920 : SUCCESS in 8m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/38a23081d2ff44eba5045df6c184765d : SUCCESS in 1h 04m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/66fd44db24164ad5b60ec04f3b582bc3 : SUCCESS in 1h 37m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a315d40f6d2d4eea8f29e41d7dc6e706 : SUCCESS in 1h 24m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06d4af7d9d074416a14e439cc5f0b01d : SUCCESS in 1h 06m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af1fbcf0eec14e01922b9100b2f1380c : SUCCESS in 1h 01m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3c8b5bd9ae44df0b65003d8e38d83fc : SUCCESS in 1h 20m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3786650a0294e9bb0a8d0a7ae1db431 : SUCCESS in 1h 59m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b8439d50f4b4e39a83bd728d490ae58 : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":21},{"id":"405b05f4c711f87037cf00a029704377fbdf45e3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-29 01:22:53.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/792462/792462/21/check-not-vote/ext-cinder-zuul/eb6a580 : FAILURE in 0s","accounts_in_message":[],"_revision_number":21},{"id":"c0cd3b35c737a2279109e012eb5837ac9be59214","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-29 01:28:49.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/62/792462/21/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/88fac91/88fac91.html : FAILURE in 2h 43m 07s","accounts_in_message":[],"_revision_number":21},{"id":"e009456d43e588146204df5d5beffbc4ef95f73c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-29 01:45:58.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/62/792462/21/upstream-check/cinder-cDOT-iSCSI/00f095f/ : SUCCESS in 33m 27s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/62/792462/21/upstream-check/cinder-cDOT-NFS/416c9ce/ : SUCCESS in 36m 26s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/62/792462/21/upstream-check/cinder-cDOT-FCP/68eb4e5/ : SUCCESS in 1h 01m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"468bf418bc929eb56ad75b06130f8c37101cfdc8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-29 01:51:05.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/62/792462/21/upstream-solidfire-check/cinder-solidfire-iSCSI/9b6b43d/ : SUCCESS in 1h 05m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":21},{"id":"da75be6deb29456ece5f57a2c940437c27505e22","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-29 02:25:48.000000000","message":"Patch Set 19:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-19/ns5_nfs : SUCCESS in 4h 24m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-19/ns5_iscsi : SUCCESS in 4h 28m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":19},{"id":"38ba4a9ae28348c53c79e524e1f8206f99b79392","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-29 03:11:56.000000000","message":"Patch Set 20:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-20/ns5_nfs : FAILURE in 53m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-20/ns5_iscsi : SUCCESS in 4h 30m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":20},{"id":"c5e760f6d67165dc0e80a1519c2e306061731650","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-29 03:58:13.000000000","message":"Patch Set 21:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-21/ns5_nfs : SUCCESS in 4h 28m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-21/ns5_iscsi : SUCCESS in 4h 31m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":21},{"id":"2f35bfb96ded17f609945802fe80d7aad762fd9a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-06-29 04:20:20.000000000","message":"Patch Set 21:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792462-21/Seagate-iSCSI : SUCCESS in 39:17.35\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792462-21/Seagate-FC : SUCCESS in 36:13.98","accounts_in_message":[],"_revision_number":21},{"id":"4c1967280448c763d89b4cc28a77c222af147876","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-29 05:00:07.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/62/792462/21/check/DellEMC_PowerFlex/03032c0/DellEMC_PowerFlex/244 : FAILURE in 2h 13m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"dc6cbebb7361298c490917f3322892962992cef2","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-29 05:06:23.000000000","message":"Patch Set 21:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792462-21/HPMSA-iSCSI : FAILURE in 58:53.62\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792462-21/HPMSA-FC : SUCCESS in 31:21.39","accounts_in_message":[],"_revision_number":21},{"id":"8d5adcf12664ce3ddbc2aa90ad1bfda9f7d28ffc","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-29 05:16:12.000000000","message":"Patch Set 21:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792462-21/Lenovo-iSCSI : SUCCESS in 46:12.48\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792462-21/Lenovo-FC : SUCCESS in 42:31.67","accounts_in_message":[],"_revision_number":21},{"id":"5bdd9e7a5c3ed971dadd00e9a9ca9bf8459773b6","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-29 06:25:01.000000000","message":"Patch Set 21:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792462-21/PVME-iSCSI : SUCCESS in 55:04.17\n* PVME-FC http://os-logs.tristero.net/dell-ci/792462-21/PVME-FC : SUCCESS in 52:15.94","accounts_in_message":[],"_revision_number":21},{"id":"d1d861f17e65a479fc7d74f5b90f3fb8078a2ac7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-29 15:28:16.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/21/20210629215306/NEC-ISCSI : SUCCESS in 2h 32m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/792462/21/20210629215306/NEC-FC : SUCCESS in 2h 35m 08s","accounts_in_message":[],"_revision_number":21},{"id":"2b611061958706ee458b632248a2b0b14fa5a6ec","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-29 15:38:35.000000000","message":"Patch Set 21: Code-Review+2\n\nRevisions look good to me.","accounts_in_message":[],"_revision_number":21},{"id":"d67aa78ce4cd66f31b69ef7dbaf9c55cc048fa49","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-29 20:37:56.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/62/792462/21/check/DellEMC_VNX_iSCSI/ecfb286/DellEMC_VNX_iSCSI/207 : FAILURE in 3h 55m 33s\n- DellEMC_VNX_FC http://publiclogs.emc.com/62/792462/21/check/DellEMC_VNX_FC/142a9e5/DellEMC_VNX_FC/223 : SUCCESS in 1h 35m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":21},{"id":"a849983085344f1a2d3eb6337fbde51db258aa98","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-30 01:54:57.000000000","message":"Patch Set 21:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-21/ns5_nfs : SUCCESS in 4h 35m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-62-792462-21/ns5_iscsi : SUCCESS in 4h 41m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":21},{"id":"77a40304384757514e7473fd3309ad9963918317","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-30 07:09:54.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":21},{"id":"c66609fa53972889e53b7a766bbf85dd1865e282","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-30 16:49:30.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)\n\nOne typo in the commit message.  Don\u0027t think this is worth blocking the patch.","accounts_in_message":[],"_revision_number":21},{"id":"3f18f021891d40c7a98b8d85d5af55b78977f5bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:49:59.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"a9cc8183d52d812435552b4b82e486601dab42ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:02:32.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6626c3238c54e339150842cdf9a5720 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c31731ba9a44935a280c32e63505e04 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7336ae9c8d14a3bbcc96feea7e9cc55 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee85475ec5348589a90d73088f2125a : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd096ce9b3f64923a7cc703d88a2b88a : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bf243e3a69464ab9bc6dc5227b1327 : SUCCESS in 13m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bb550df67b64c55a0efdcd4082c4e49 : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/a886fcf52fa94e1abb726bdf8b658dd1 : SUCCESS in 1h 03m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab7fb6389de94c6896e8b2d9be2af73d : SUCCESS in 1h 00m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8592c8545e420d8164efb7809ac9df : SUCCESS in 7m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/37752f0a915e4858bfeede2581e2bdb4 : SUCCESS in 1h 10m 56s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/642d48dc40db431b898d6948197853e9 : SUCCESS in 1h 10m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba7d23dbf49247da8ae7a509f3fd0f17 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":21},{"id":"ec667a8a552ff2707ea0cb7cf47e4256ee541462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:02:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":21},{"id":"113c17f77394fd3919712b996bfd31d18e260cca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:04:21.000000000","message":"Patch Set 21:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba84bf1595742a1bf8f61e33423ecee : FAILURE in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b35e27a951b647ea94a10a2bb41686d0 : FAILURE in 58s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"49848a7674b3a502e7cf6d610f0dc5d2dddbd642","revisions":{"1e862b964c2795ad3d81e8df5a5c2923d64627b1":{"kind":"REWORK","_number":1,"created":"2021-05-20 17:37:41.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/1"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"subject":"Fix minimum SDK version required","message":"Fix minimum SDK version required\n\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e862b964c2795ad3d81e8df5a5c2923d64627b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e862b964c2795ad3d81e8df5a5c2923d64627b1"}]},"branch":"refs/heads/master"},"1e3ac4135ef6def88cea6f1d67c63f4c8087b9b9":{"kind":"REWORK","_number":2,"created":"2021-05-20 19:01:26.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/2"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 19:00:54.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e3ac4135ef6def88cea6f1d67c63f4c8087b9b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e3ac4135ef6def88cea6f1d67c63f4c8087b9b9"}]},"branch":"refs/heads/master"},"5bb3e8dae9bf38da23191a9c7635f0e6129cf17b":{"kind":"REWORK","_number":3,"created":"2021-05-20 20:22:31.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/3"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 20:22:05.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bb3e8dae9bf38da23191a9c7635f0e6129cf17b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bb3e8dae9bf38da23191a9c7635f0e6129cf17b"}]},"branch":"refs/heads/master"},"5c9acadb129f499a266c86657ad76903f5b47af2":{"kind":"REWORK","_number":4,"created":"2021-05-20 21:02:14.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/4"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 21:01:50.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c9acadb129f499a266c86657ad76903f5b47af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c9acadb129f499a266c86657ad76903f5b47af2"}]},"branch":"refs/heads/master"},"b6a5e5da34a7584374f4cf95f5c02c9e5cef6903":{"kind":"REWORK","_number":5,"created":"2021-05-21 13:51:08.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/5"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-21 13:40:26.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST versin on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6a5e5da34a7584374f4cf95f5c02c9e5cef6903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6a5e5da34a7584374f4cf95f5c02c9e5cef6903"}]},"branch":"refs/heads/master"},"d169f2934e2793e3055426215d7c7923d4bfee29":{"kind":"REWORK","_number":6,"created":"2021-05-21 14:12:30.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/6"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-21 14:11:50.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST versin on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d169f2934e2793e3055426215d7c7923d4bfee29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d169f2934e2793e3055426215d7c7923d4bfee29"}]},"branch":"refs/heads/master"},"294fa17066d1e263cf12f011ef601b236e0c5535":{"kind":"REWORK","_number":7,"created":"2021-05-21 14:36:59.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/7"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-21 14:36:38.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST versin on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/294fa17066d1e263cf12f011ef601b236e0c5535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/294fa17066d1e263cf12f011ef601b236e0c5535"}]},"branch":"refs/heads/master"},"b44ba7bfe554176d4363ae9f70d28153203d5264":{"kind":"REWORK","_number":8,"created":"2021-05-21 15:01:00.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/8"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-21 15:00:27.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST versin on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b44ba7bfe554176d4363ae9f70d28153203d5264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b44ba7bfe554176d4363ae9f70d28153203d5264"}]},"branch":"refs/heads/master"},"3166900303fc95fea30ffe450af0fff5a2171c71":{"kind":"REWORK","_number":9,"created":"2021-06-17 19:24:24.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/9"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-17 19:22:41.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3166900303fc95fea30ffe450af0fff5a2171c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3166900303fc95fea30ffe450af0fff5a2171c71"}]},"branch":"refs/heads/master"},"8c22f6e79cc0810024cf30fed80b022cc3e49851":{"kind":"REWORK","_number":10,"created":"2021-06-18 15:42:18.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/10"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-18 15:41:45.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c22f6e79cc0810024cf30fed80b022cc3e49851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c22f6e79cc0810024cf30fed80b022cc3e49851"}]},"branch":"refs/heads/master"},"1cca483fdc302fe8b38e797cdfa556f04eb04ed0":{"kind":"REWORK","_number":11,"created":"2021-06-18 15:43:17.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/11"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-18 15:42:58.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cca483fdc302fe8b38e797cdfa556f04eb04ed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cca483fdc302fe8b38e797cdfa556f04eb04ed0"}]},"branch":"refs/heads/master"},"2d8f2a1fe75fb169a56f60b6ef9d0a1bebd49a83":{"kind":"REWORK","_number":12,"created":"2021-06-18 15:47:28.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/12"}}},"commit":{"parents":[{"commit":"87059642f470264c57f4d03adcc99c84b31bccd5","subject":"Merge \"Tests: test_backup - specify volume_type_id for snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87059642f470264c57f4d03adcc99c84b31bccd5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-18 15:47:03.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d8f2a1fe75fb169a56f60b6ef9d0a1bebd49a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d8f2a1fe75fb169a56f60b6ef9d0a1bebd49a83"}]},"branch":"refs/heads/master"},"e2cadc4b453522a33f6c13584cf06f96125e32d9":{"kind":"REWORK","_number":13,"created":"2021-06-21 15:17:41.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/13"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-21 15:09:29.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2cadc4b453522a33f6c13584cf06f96125e32d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2cadc4b453522a33f6c13584cf06f96125e32d9"}]},"branch":"refs/heads/master"},"32af0ecd4d2574dfe191bf8de181c36a8fed6c6f":{"kind":"REWORK","_number":14,"created":"2021-06-21 16:13:36.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/14"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-21 16:13:09.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/32af0ecd4d2574dfe191bf8de181c36a8fed6c6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/32af0ecd4d2574dfe191bf8de181c36a8fed6c6f"}]},"branch":"refs/heads/master"},"96f41218540eb3ca0e2787d23b1422443b5d3e1f":{"kind":"REWORK","_number":15,"created":"2021-06-21 16:41:52.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/15"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-21 16:41:14.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nNote - there are a lot of formatting changes as `black` has been\nrun against this code for readability purposes.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96f41218540eb3ca0e2787d23b1422443b5d3e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96f41218540eb3ca0e2787d23b1422443b5d3e1f"}]},"branch":"refs/heads/master"},"33d83438e749d96d8396df47967147cc52451e63":{"kind":"REWORK","_number":16,"created":"2021-06-23 20:14:35.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/16"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-23 20:13:55.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33d83438e749d96d8396df47967147cc52451e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33d83438e749d96d8396df47967147cc52451e63"}]},"branch":"refs/heads/master"},"25f7485ece0dd6b8443e770cb48842b7dc202ad1":{"kind":"REWORK","_number":17,"created":"2021-06-23 20:17:59.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/17"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-23 20:16:56.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25f7485ece0dd6b8443e770cb48842b7dc202ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25f7485ece0dd6b8443e770cb48842b7dc202ad1"}]},"branch":"refs/heads/master"},"9ec8e1654c7ccbd1c48cd44d57aa36b202af06bd":{"kind":"REWORK","_number":18,"created":"2021-06-23 20:34:15.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/18"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-23 20:33:49.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ec8e1654c7ccbd1c48cd44d57aa36b202af06bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ec8e1654c7ccbd1c48cd44d57aa36b202af06bd"}]},"branch":"refs/heads/master"},"266e921161569187b9a5d40ab813ff9dca8506c1":{"kind":"REWORK","_number":19,"created":"2021-06-28 21:42:49.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/19"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-28 21:42:24.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/266e921161569187b9a5d40ab813ff9dca8506c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/266e921161569187b9a5d40ab813ff9dca8506c1"}]},"branch":"refs/heads/master"},"4735f8bca33ba3aa08252ced92c6b47f91cd3b6c":{"kind":"REWORK","_number":20,"created":"2021-06-28 21:45:20.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/20"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-28 21:44:43.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4735f8bca33ba3aa08252ced92c6b47f91cd3b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4735f8bca33ba3aa08252ced92c6b47f91cd3b6c"}]},"branch":"refs/heads/master"},"49848a7674b3a502e7cf6d610f0dc5d2dddbd642":{"kind":"REWORK","_number":21,"created":"2021-06-28 22:41:01.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/62/792462/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/792462/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/792462/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/792462/21"}}},"commit":{"parents":[{"commit":"0caaa1fee5951c4dbd40f71639d611f9cf94f5f7","subject":"Merge \"[Pure Storage] Ensure multiattach volumes are not disconnected early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0caaa1fee5951c4dbd40f71639d611f9cf94f5f7"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-05-20 17:36:58.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2021-06-28 22:40:25.000000000","tz":-240},"subject":"[Pure Storage] Fix minimum SDK version required","message":"[Pure Storage] Fix minimum SDK version required\n\nChange method to check supported REST version on the backend.\nThis will mkae it easier to specify minimum REST versions for\nspecific feature support.\n\nCloses-Bug: #1929219\nChange-Id: Ibab932d3536d44dd7414560768ed5bdeaa9e61e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49848a7674b3a502e7cf6d610f0dc5d2dddbd642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49848a7674b3a502e7cf6d610f0dc5d2dddbd642"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
