)]}'
{"id":"openstack%2Fcinder~792560","triplet_id":"openstack%2Fcinder~master~Ibae773dd3bd185f74c37709ee396049d2b627b74","project":"openstack/cinder","branch":"master","topic":"fix-vmware-bfv","hashtags":[],"change_id":"Ibae773dd3bd185f74c37709ee396049d2b627b74","subject":"Fix the issue of creating volume from image for vmware","status":"ABANDONED","created":"2021-05-21 09:50:40.000000000","updated":"2025-12-12 21:32:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1cb1356ee2652120323e8130294abc9f147670f3","_number":792560,"virtual_id_number":792560,"owner":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-21 12:50:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 10:00:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 10:18:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 10:19:43.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 10:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 10:46:12.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-21 11:12:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-21 11:34:08.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 12:22:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-21 12:50:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 12:57:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-21 13:11:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 14:33:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-22 12:04:04.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-24 07:18:31.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 11:53:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b84f76cb981bdc5d97afdd1eb19c61ded388fd2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"date":"2021-05-21 09:50:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a32a992cd5fa42f952e4c3933133a2aa4dffd3c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-21 10:00:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/60/792560/1/check/cinder-storpool-tempest/40c2692/ : FAILURE in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1886e8dec2c707f22249875508317d3f01c4906d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 10:18:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/792560/1/check/iscsi-driver-nimble/49fdb7a/49fdb7a.html : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a53e795773e481d06626e25a7e9c499fa1dd7010","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-21 10:19:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/792560/1/ : FAILURE in 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"73cd9e8441dfe3942c7a64bfbb90ac88832d7440","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:23:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bcf491310c5494f986fbb1d71ef8ce9 : FAILURE in 28m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20de3459d4e34b3483ab8339f74f4411 : FAILURE in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5467525d0ea06cc9f140ac93ae8de88badfaad8f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-21 10:46:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792560/1/iscsi/ : SUCCESS in 53m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/792560/1/smb/ : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9be922fedd841a26e75e73801d58a3496ac374b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 11:00:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/792560/1/check/iscsi-driver-primera/b97cae0/b97cae0.html : FAILURE in 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1ce1e38c422c4092e00eb2063b28c9571b20f904","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-21 11:12:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/792560/1/check/dsvm-tempest-cisco-zonemanager-job/60e4be1/ : FAILURE in 1h 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bc44c7220ee2947c769f8aff910f1d5eebdeb63a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 11:34:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/60/792560/1/check/DellEMC_XtremIO_FC/71f42fc/DellEMC_XtremIO_FC/188 : SUCCESS in 1h 01m 53s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/60/792560/1/check/DellEMC_XtremIO_iSCSI/68ccdb0/DellEMC_XtremIO_iSCSI/195 : FAILURE in 1h 42m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd5e09291cea8ae81fd88f0ea3de49cbd9c1899d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 12:11:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/792560/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3dc081a/3dc081a.html : SUCCESS in 2h 19m 15s","accounts_in_message":[],"_revision_number":1},{"id":"eff62e4787628f240fbed572c049353987b93022","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-21 12:22:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/60/792560/1/check/DellEMC_Unity_FC/cf01903/DellEMC_Unity_FC/201 : SUCCESS in 1h 22m 35s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/60/792560/1/check/DellEMC_Unity_iSCSI/a6e29bd/DellEMC_Unity_iSCSI/187 : FAILURE in 1h 38m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"83074075c96d7dfac5c3f81c91f213d5a4f0f3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:50:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b9acf71a8cf4fbdad7b86d4bf312a3a : FAILURE in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/846ac41cf8444226a6ff0eab9638101c : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac2cced3dfd40a39c508108cbd8b02b : FAILURE in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36b24ae3a4094afa9dee06e903cec9ca : FAILURE in 12m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50d1692e49ab4500b85882952844b1f6 : FAILURE in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66dc31fd8a2a45eb917dce8e6d56f68e : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a45e073dc6864366b8f35c06da222281 : SUCCESS in 1h 01m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d55f15ef04c947eda6a4333688b7b036 : SUCCESS in 1h 14m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/57855055dc4a4324adeeca6dfe628b8b : FAILURE in 13m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a05e946f2c3d4d01b0661f5b84eb099b : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b83c1bbc1f64b43989a6b2bf1590bb5 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ea4e7b87fd04d27925a5612e0f595ec : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8fdbfb50934f49dd8f7e3afdbe36a02b : FAILURE in 36m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7213f5067b7041dd951e5b784e81ad58 : FAILURE in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b5bc22693394ad3b8719340fceb8ee8 : TIMED_OUT in 2h 06m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ac2878f9372d403c8a9d75a245417420 : SUCCESS in 1h 15m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d15e6f251864707a8b142b939388a55 : SUCCESS in 1h 30m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/21998ccb854f40739f3027e3c3eec6a1 : SUCCESS in 1h 17m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d709c63e8094387bdba9e75fc1de29b : FAILURE in 1h 23m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef959eeddb684927bca017cdea93efc8 : SUCCESS in 1h 12m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd4b1c27c7aa401c812df39bafa0f3d0 : SUCCESS in 2h 09m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a2efb3fd4d647ddbb75d276e3545f46 : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"db75e8f4a9d8cff58795eb1b4974720f258cbc81","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-21 12:57:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/792560/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/40b8a35/ : FAILURE in 1h 45m 47s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/792560/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cab92f2/ : TIMED_OUT in 3h 05m 11s","accounts_in_message":[],"_revision_number":1},{"id":"460ff1da8895b10570d0cfd2010a146a5627dfa5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-21 13:11:32.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/0ba48b601aa74c4fb3a74eb08447ccfb : POST_FAILURE in 1h 40m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4fb26a027c4d47f086281e79fffcad0a : POST_FAILURE in 1h 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"91a4d57cf105ddf6ac7dfd53a1fb27ba552cb6d8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 14:33:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-792560-1/ns5_nfs : SUCCESS in 4h 5m 56s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-792560-1/ns5_iscsi : SUCCESS in 4h 10m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8d8a8ab9561e280093cde18d169f9edf5a3858ce","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-22 12:04:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/60/792560/1/check/DellEMC_PowerMAX_FC/7e1a832/DellEMC_PowerMAX_FC/90 : SUCCESS in 5h 27m 09s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/60/792560/1/check/DellEMC_PowerMAX_iSCSI/5256197/DellEMC_PowerMAX_iSCSI/80 : SUCCESS in 5h 08m 30s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e694d3bba1d1c9b6b18521fcbad819d5ad1d9e74","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-24 07:18:31.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/792560/1/20210524111651/NEC-ISCSI : SUCCESS in 2h 35m 39s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/792560/1/20210524111651/NEC-FC : FAILURE in 5h 01m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c87d866fe6dd99644fbd10a80d9ef3eb355be1d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-25 11:53:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/792560/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1cb1356ee2652120323e8130294abc9f147670f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 21:32:17.000000000","message":"Abandoned\n\nVery old at this point, so closing this out. Feel free to reopen and update or submit something new, if still needed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5836e8f85a27c69c3c57dd6ed87a5ac02af2cf45","revisions":{"5836e8f85a27c69c3c57dd6ed87a5ac02af2cf45":{"kind":"REWORK","_number":1,"created":"2021-05-21 09:50:40.000000000","uploader":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"ref":"refs/changes/60/792560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/792560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/792560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/792560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/792560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/792560/1"}}},"commit":{"parents":[{"commit":"9d504800f0c4c17ad919351eca1e00332f29737b","subject":"Merge \"Creating snapshot on NFS backend fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d504800f0c4c17ad919351eca1e00332f29737b"}]}],"author":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-21 09:37:42.000000000","tz":480},"committer":{"name":"ZijianGuo","email":"guozijn@gmail.com","date":"2021-05-21 09:50:31.000000000","tz":480},"subject":"Fix the issue of creating volume from image for vmware","message":"Fix the issue of creating volume from image for vmware\n\nIn the vmware environment, if we create a volume from an image, there is no need\n to create an empty volume, otherwise an error will be reported that the target\nvolume already exists.\n\nSigned-off-by: ZijianGuo \u003cguozijn@gmail.com\u003e\nChange-Id: Ibae773dd3bd185f74c37709ee396049d2b627b74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5836e8f85a27c69c3c57dd6ed87a5ac02af2cf45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5836e8f85a27c69c3c57dd6ed87a5ac02af2cf45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
