)]}'
{"id":"openstack%2Fcinder~792642","triplet_id":"openstack%2Fcinder~master~I7aa9d603f69ba4424da7ae6f98192a888da1071c","project":"openstack/cinder","branch":"master","topic":"mypy","hashtags":[],"change_id":"I7aa9d603f69ba4424da7ae6f98192a888da1071c","subject":"Run cinder-mypy job in the check","status":"MERGED","created":"2021-05-21 14:46:28.000000000","updated":"2021-08-05 20:23:45.000000000","submitted":"2021-08-04 17:25:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792642-mypy","meta_rev_id":"219620c6a8285b578f901050df0e0fd55abe14b1","_number":792642,"virtual_id_number":792642,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-04 17:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-03 14:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-08-02 01:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-08-05 20:23:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2021-07-16 14:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-16 15:22:25.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2021-08-05 20:23:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-08-03 14:01:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"}],"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 15:07:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 15:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 16:00:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-21 16:08:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 16:08:34.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 16:23:57.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-21 16:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 17:00:59.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-21 17:03:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-21 17:07:25.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 19:37:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 21:07:49.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-05-21 21:29:33.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 21:30:05.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-05-21 21:56:40.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-05-21 22:16:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-21 22:28:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 23:30:28.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-05-22 03:22:14.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-05-22 06:54:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-23 02:36:39.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-24 17:23:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-18 14:31:30.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-06-22 14:02:16.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-06-30 09:24:54.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-07-09 21:40:15.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-07-10 00:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-07-12 07:40:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 14:19:46.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 14:33:48.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-07-19 02:19:10.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2021-08-02 01:07:56.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-08-03 14:01:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"b572e88c41ab7d523b2f1b9ef8e3c6e3d4e6f9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-21 14:46:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fda04debda4b62773866f4c0aa6c1fae75ecd6be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-21 15:07:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/42/792642/1/check/cinder-storpool-tempest/b0dfc59/ : FAILURE in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"65713527e0f0d7cc302ec00bf5e2c560cae8a8c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:10:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/618f8c1c7ed44bc7b96b4503a32a2dd6 : SUCCESS in 17m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c30ca4bd2ce49c6b1d7ff9571fabafd : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0795c5d6fa3d3282222ffeb240c54b78fdeff679","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-21 16:00:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/1/iscsi/ : SUCCESS in 1h 12m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/1/smb/ : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cb825edfc0e5323159babd450966723a0ae612ba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 16:08:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/792642/1/check/iscsi-driver-primera/4a4583e/4a4583e.html : FAILURE in 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7ef565844c9497c2982d7fe5ccd91750b50756f0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-05-21 16:08:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/792642/1/ : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4002c9caeb747d8ca7e9ee7c616a650c2b07a75e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 16:15:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/42/792642/1/check/iscsi-driver-nimble/29c0b22/29c0b22.html : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"774d0288b7b6c3f4b787aa8600eae92afe52236b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-21 16:23:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/42/792642/1/check/dsvm-tempest-cisco-zonemanager-job/34e4c00/ : FAILURE in 1h 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4622b7cbba31c8fcb0cf6713c0aef18885319e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 16:48:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b98ca9d16e94c309d0953d84b4d3d5e : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76de1d834b994bf8a06f610df31fd5f7 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781ff705778644a7a9e300466334c72a : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b94d3d73eb6144c59bc9b92be23f09d3 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2da96403a6a43c7b40d10c934f254ca : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a92bc8ec0048ebafd46d3a15127f9c : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/f75f2a6c12ca4ea8a4f9903b00cbca03 : SUCCESS in 58m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20db952f10144cf8a4c95d5e3add3c8f : SUCCESS in 1h 17m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b93e1ce2cfe84bddbfd90d6c0acb1ebd : SUCCESS in 15m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/45c2fab8772a4cb18250aaa183c4ac41 : SUCCESS in 5m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a0e3074fb1c749a0bb3542ec55cd5433 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef5b020320c14199a31d6745aa501786 : SUCCESS in 6m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f6b442bc65c34d1d9e50975c96a9e2c6 : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad5a4eacfaad486e93f98a0a2cb7467d : FAILURE in 34m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1182391a556f42dfb68f5511d59d8c46 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9a42ab53a514bb099bb53999f633639 : SUCCESS in 1h 24m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08e28abcaacd4590ae88ab6b1d067aeb : SUCCESS in 1h 31m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/712fc9260f04420bb29c2d80f6878baf : SUCCESS in 1h 28m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8125a0694729462ca6165320def3be69 : SUCCESS in 1h 12m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2802e02d58e84738befbce5392219513 : SUCCESS in 1h 10m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/124b844d23f44b138f4594598f625524 : SUCCESS in 1h 17m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc1a381e5fc1407b927c3132cf5d455e : SUCCESS in 1h 57m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89afba625354458db8f523f5cfa24b2a : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"16b6003ead853aeee27207226b608bf39483a2e2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-21 17:00:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/42/792642/1/upstream-solidfire-check/cinder-solidfire-iSCSI/dae0536/ : FAILURE in 11m 37s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"5de73305643edb79f6a4c564189aa585a336e733","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-21 17:03:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/792642/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/14cd70a/ : POST_FAILURE in 2h 16m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/792642/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a6909e4/ : SUCCESS in 2h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f8defa59342fb65816d78d114d4d3c83c062cc75","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-21 17:07:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/42/792642/1/upstream-check/cinder-cDOT-iSCSI/fc0e3fe/ : FAILURE in 11m 25s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/42/792642/1/upstream-check/cinder-cDOT-NFS/4c01564/ : FAILURE in 10m 30s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/42/792642/1/upstream-check/cinder-cDOT-FCP/108cc20/ : FAILURE in 18m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"25232fda0dc8ba7b4057aa46a1be5dc230c0ccfc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 17:45:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/42/792642/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/27be5a1/27be5a1.html : SUCCESS in 2h 36m 50s","accounts_in_message":[],"_revision_number":1},{"id":"12cd4c29340ce4f126f568004cae84f304f9cd6c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-21 19:37:48.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7fc4bd15645246a09e57243302bb92f7 : POST_FAILURE in 1h 32m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e9ed8acf78964948b4d2612fa81bad99 : POST_FAILURE in 1h 23m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3e34c921b51fb9d71ef9dd5636693b100266a54a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-05-21 21:07:49.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792642-1/Seagate-iSCSI : SUCCESS in 46:46.40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792642-1/Seagate-FC : SUCCESS in 38:56.56","accounts_in_message":[],"_revision_number":1},{"id":"bfccc3df0ee671cde54b0ebcf5f86241883cf0f7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-21 21:29:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/42/792642/1/check/DellEMC_XtremIO_FC/c406007/DellEMC_XtremIO_FC/194 : FAILURE in 1h 19m 15s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/42/792642/1/check/DellEMC_XtremIO_iSCSI/17b8a80/DellEMC_XtremIO_iSCSI/199 : SUCCESS in 1h 52m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"c70bf194b32714592dc73c47fc4365c6584ebd8e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-21 21:30:05.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/42/792642/1/check/DellEMC_PowerStore_iSCSI/3bb54bc/DellEMC_PowerStore_iSCSI/116 : SUCCESS in 53m 03s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/42/792642/1/check/DellEMC_PowerStore_FC/a16cc08/DellEMC_PowerStore_FC/117 : SUCCESS in 56m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"871fc45391150cf9cd4f7433fc1f7e1dc9b891ba","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-05-21 21:56:40.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792642-1/Lenovo-iSCSI : SUCCESS in 52:11.66\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792642-1/Lenovo-FC : SUCCESS in 46:36.85","accounts_in_message":[],"_revision_number":1},{"id":"87c391652533b47336df57854a071d82cfaefb6d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-21 22:16:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/42/792642/1/check/DellEMC_Unity_FC/6eb14b5/DellEMC_Unity_FC/205 : SUCCESS in 1h 39m 18s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/42/792642/1/check/DellEMC_Unity_iSCSI/cc31616/DellEMC_Unity_iSCSI/191 : SUCCESS in 2h 03m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4990afdfcc98c03121cecaf1245bbe2911fbbb2a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-21 22:28:34.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-1/ns5_nfs : SUCCESS in 4h 51m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-1/ns5_iscsi : FAILURE in 56m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2434b60810787bd7cd652ed1dd2bee6bab32faf","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-05-21 23:30:28.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792642-1/HPMSA-iSCSI : FAILURE in 1:27:04\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792642-1/HPMSA-FC : SUCCESS in 33:58.25","accounts_in_message":[],"_revision_number":1},{"id":"9225f167e1127d766cd50ef732d3ac39ea4d67ee","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-22 03:22:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/42/792642/1/check/DellEMC_SC_FC/baf058c/DellEMC_SC_FC/132 : SUCCESS in 1h 07m 33s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/42/792642/1/check/DellEMC_SC_iSCSI/7453b07/DellEMC_SC_iSCSI/129 : SUCCESS in 2h 09m 41s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"adcb045f7b01cf1dfa5c0a0eb96350e722f9994d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-22 03:23:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-1/ns5_nfs : SUCCESS in 4h 22m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-1/ns5_iscsi : SUCCESS in 4h 26m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a90e92a44782f67f0c45c9844ac1602a33be724a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-22 06:54:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/42/792642/1/check/DellEMC_VNX_iSCSI/34619a8/DellEMC_VNX_iSCSI/133 : FAILURE in 1h 11m 50s\n- DellEMC_VNX_FC http://publiclogs.emc.com/42/792642/1/check/DellEMC_VNX_FC/c890ca2/DellEMC_VNX_FC/138 : SUCCESS in 1h 40m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"804fe260f995110577865bc07fc4821e8c439623","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-23 02:36:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/42/792642/1/check/DellEMC_PowerMAX_FC/fdbbac2/DellEMC_PowerMAX_FC/93 : SUCCESS in 4h 41m 44s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/42/792642/1/check/DellEMC_PowerMAX_iSCSI/b90c189/DellEMC_PowerMAX_iSCSI/83 : SUCCESS in 5h 02m 06s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"28784c3d87ba5e9f1b995dbb89688640e38e256a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-24 17:23:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/1/20210524234841/NEC-FC : SUCCESS in 2h 24m 12s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/1/20210524234841/NEC-ISCSI : SUCCESS in 2h 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bab634334a010c4fe74164112a3c6aba4c9bb859","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-31 05:49:50.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-1/ns5_nfs : SUCCESS in 4h 58m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-1/ns5_iscsi : SUCCESS in 5h 5m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e3fea77c90ced9e72fc7afe5a3afb6d459f8de14","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-06-18 14:31:30.000000000","message":"Patch Set 1: Code-Review+1\n\nagree with decision.","accounts_in_message":[],"_revision_number":1},{"id":"f685eb6331fbfd103abd37111b33c768a64909bf","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-22 14:02:16.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"9d3f64beb3a01820f0285f644cc0491fed13271d","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-06-30 09:24:54.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"1de46769672d0b6644224ff8c8ec53bb166982bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-09 17:20:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"69156e84c1bc72d614556e46ea6840c3f8ab70bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 17:40:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63ddf0df6c20481483bdb34dfcc639cb : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/534d3a5950df4f0fbde5899d938d92e6 : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9203e7b1a2f229746a2904d34748046157a4cb2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-09 18:16:24.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"1cad3cec2d371b590b3aac42ef4af41a6256e70d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 18:24:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/792642/2/check/iscsi-driver-primera/4ce6cd4/4ce6cd4.html : FAILURE in 1h 02m 00s","accounts_in_message":[],"_revision_number":2},{"id":"22ff5d80584b1e5f93d3b6ee6b3ec0dea67a4015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 19:43:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb034c261504d1db26e2af7481a8904 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95624f30969f4c6c8816ed7a04034a58 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e3cfc4de8ac487db02d20aece39b4f0 : SUCCESS in 10m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfbb3ccd99114e82bcc29857b4201064 : SUCCESS in 9m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6688deb404af4c3b9b867ca8991bf571 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fe8d1c598c409e9cceb877c2f5aa85 : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/9491685c6271474fa775d0729eceeb2d : SUCCESS in 56m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf4e914e4ab94f83ac80a829e132fe9d : SUCCESS in 1h 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/228c9f0c7bb74299b552cff2d0d1ec85 : SUCCESS in 12m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b66ee6c21e594940941585b0e9e5082a : FAILURE in 6m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36473d87250a4ed4903434aeea77858a : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b924dd3912cc434796de1dfa8ae652df : SUCCESS in 7m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51170386314246f59218c322b68f84ed : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5aaae64eb9f344da9a43a8f5a8799cf2 : FAILURE in 41m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c98cde67ac17426ba53122f71db5b22a : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a64bc0e50ef942778c225e4048a0a056 : SUCCESS in 1h 13m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee242a6ef6ec4c00a5c66ec8b7c10e5d : SUCCESS in 1h 25m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19f011355cd245d9897b93acafed4f74 : SUCCESS in 1h 34m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9de585b4dd04dddbf393cef103567b2 : SUCCESS in 1h 08m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58b8cd4f5e57410d8318709e85d00111 : SUCCESS in 1h 25m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43f4d95246174ed5a41d656ea421b48b : SUCCESS in 1h 15m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb5b44bbd08c4291b6a98989dc869461 : SUCCESS in 2h 15m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a075db749f614a73b42962658bf44c28 : SUCCESS in 46m 20s\n\nWarning:\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":2},{"id":"def0a5a5104420199b7716c4561e7fc62e625f57","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 19:44:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/42/792642/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9190ea1/9190ea1.html : SUCCESS in 2h 22m 32s","accounts_in_message":[],"_revision_number":2},{"id":"20813db3457e82dda319d4544c34bf6dd81e7e13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-09 20:05:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d0f40d4cb6bce5ef121885252e1d7adb2a5dd144","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-09 20:06:31.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"37f380fec407c01d9a07533dd422b717c665d67c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 21:09:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/792642/3/check/iscsi-driver-primera/b07205d/b07205d.html : FAILURE in 1h 01m 38s","accounts_in_message":[],"_revision_number":3},{"id":"f66f1ca7557eb950cfdc2cdcf8c504241745fa75","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-09 21:10:34.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/3/iscsi/ : FAILURE in 1h 04m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/3/smb/ : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"49cc17607f2e959e8d392aeee509e59b68e3a23b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-09 21:30:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/792642/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b0442e3/ : SUCCESS in 1h 14m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/792642/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f28521b/ : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8899117377610f15e6e3004c9b592fedeb7204b6","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-09 21:40:15.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792642-2/PVME-iSCSI : FAILURE in 9:17.75\n* PVME-FC http://os-logs.tristero.net/dell-ci/792642-2/PVME-FC : SUCCESS in 47:37.55","accounts_in_message":[],"_revision_number":2},{"id":"f7a42a3096557a785b081c546eff88df113d6706","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 22:13:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/42/792642/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4b0d43c/4b0d43c.html : SUCCESS in 2h 06m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0a66c8c5086d461ec6239b2ab4fd49fe46c1842b","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-09 22:59:15.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792642-2/Seagate-iSCSI : SUCCESS in 43:26.99\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792642-2/Seagate-FC : SUCCESS in 40:35.70","accounts_in_message":[],"_revision_number":2},{"id":"776dc27f0ea23ecc92511912af214282a7733300","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-09 23:45:51.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792642-2/Lenovo-iSCSI : SUCCESS in 50:07.05\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792642-2/Lenovo-FC : SUCCESS in 46:35.38","accounts_in_message":[],"_revision_number":2},{"id":"9e72dd36d46c01d86bdbd0c05325e13c547212c3","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-09 23:48:47.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792642-2/HPMSA-iSCSI : FAILURE in 1:01:55\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792642-2/HPMSA-FC : SUCCESS in 34:04.57","accounts_in_message":[],"_revision_number":2},{"id":"2354e3fdc087f393fb032f0698426e2ab6b54745","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-10 00:00:05.000000000","message":"Patch Set 2:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/792642/2/fc : SUCCESS in 0h 45m 22s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/792642/2/iscsi : SUCCESS in 1h 31m 28s","accounts_in_message":[],"_revision_number":2},{"id":"67aff61537bbb4b5d04d2111a5f8d3c9057d0d8b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-10 00:07:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/42/792642/3/check/dsvm-tempest-cisco-zonemanager-job/780afb6/ : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":3},{"id":"2919694d2aa41439de95280eeee9eb8ed6285872","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-10 00:42:45.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-2/ns5_nfs : SUCCESS in 4h 50m 33s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-2/ns5_iscsi : SUCCESS in 4h 53m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"ba738a26cc6ae35f3f9e57558026e868d52af882","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-10 03:30:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/42/792642/3/check/DellEMC_Unity_FC/bf6c883/DellEMC_Unity_FC/413 : SUCCESS in 1h 20m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/42/792642/3/check/DellEMC_Unity_iSCSI/298631c/DellEMC_Unity_iSCSI/388 : SUCCESS in 1h 43m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ee74d59d319ee7d562a2a79f6ddc878f32e09cfe","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-10 03:54:50.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-3/ns5_nfs : SUCCESS in 4h 40m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-3/ns5_iscsi : SUCCESS in 4h 43m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"9c87510ab2a455bb2599b5b44ec5c518a2cb8ef2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-10 04:15:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/42/792642/3/check/DellEMC_XtremIO_FC/0614545/DellEMC_XtremIO_FC/481 : FAILURE in 1h 20m 38s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/42/792642/3/check/DellEMC_XtremIO_iSCSI/4e7f0d1/DellEMC_XtremIO_iSCSI/471 : SUCCESS in 1h 48m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"6baf3d2a96fd3ebe952518a728ecb2db1e3d1df3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-10 17:17:36.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/42/792642/3/check/DellEMC_PowerMAX_FC/fadae81/DellEMC_PowerMAX_FC/241 : FAILURE in 5h 58m 56s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/42/792642/3/check/DellEMC_PowerMAX_iSCSI/55c836d/DellEMC_PowerMAX_iSCSI/225 : SUCCESS in 4h 17m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"5aad3a99d22079e47e5e421d3d1493b6bac0a238","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-10 22:54:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/3/20210711024912/NEC-FC : SUCCESS in 2h 32m 05s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/3/20210711024912/NEC-ISCSI : SUCCESS in 5h 05m 23s","accounts_in_message":[],"_revision_number":3},{"id":"136bbd86df98be694c6e843e01271948524f85fa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-12 07:40:37.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792642/792642/3/check-not-vote/ext-cinder-zuul/83430b3 : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":3},{"id":"bf9a07275b1cb8cbf0f9c9e9296a3666287e1aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-12 14:11:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6d7929e9dd531ac93b3fb52082e02e44192a9cfe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-12 14:30:07.000000000","message":"Patch Set 4:\n\ncinder-mypy is now passing ... not sure what happened on the last PS","accounts_in_message":[],"_revision_number":4},{"id":"9d4ea8a99e432f4e742c8fde92fee5757b431b22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:35:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d13fae2e289a4420966190d6dbba83f1 : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ada19c976614012af0009b618cce4d9 : SUCCESS in 21m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"588d86e3c420a07c02b901f91d492b559517a7cc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-12 15:16:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/792642/4/check/iscsi-driver-primera/ddc9b1a/ddc9b1a.html : FAILURE in 1h 02m 45s","accounts_in_message":[],"_revision_number":4},{"id":"34e1dc09611bc76ae2f20206aaa4d93981dfab1f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-12 16:20:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/4/iscsi/ : SUCCESS in 1h 00m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/4/smb/ : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":4},{"id":"4557a1cb9f47693344e425963b9d0f9ce477e0a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 16:27:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d5583f1394c445c802d4678df40f2f0 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde85cfc42254ecb903299be41819486 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fcf16893e884bf3817f5cfe87b4c60d : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84e6b66dbedf49479e560a025424889e : SUCCESS in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36299def888140c5b86b9988d2fb1380 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7249686052aa491ab20f42a944eebf49 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/efc61f8deb0e42a3aa8fc39c50991efd : SUCCESS in 1h 09m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0a89d26d000466888d52c1e15769221 : SUCCESS in 1h 25m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da701feda9b345d88c13de79328c5954 : SUCCESS in 14m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/52f47280d42341bd8c60a7687b96b46e : SUCCESS in 5m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e812db5b51814966b328667e04f109aa : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f75283e04e94ef797e740bcb15535dc : SUCCESS in 6m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc498be3c829462d8bd3f3e87a798d25 : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1ab0d8f554dd4367bc8542a9cdbdbfdd : FAILURE in 35m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f948292e1ed41e280e776123ebc5fe5 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/284fa234944e47cda95833d67daf9a3f : SUCCESS in 1h 32m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e49505beb9484b2796d7c094fe3e0128 : SUCCESS in 1h 19m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6365f36a1444e42a9791c960cfbdadb : SUCCESS in 1h 40m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7639477f10824d6fb9140242b3be3d57 : SUCCESS in 1h 21m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df87d31d96ff43a1b47f96215d173f94 : SUCCESS in 1h 09m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/401e74aaa9a6424cbff6bca7007ff7e8 : SUCCESS in 1h 11m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11a82fd7df9745398e70dce0d0f999ad : FAILURE in 2h 08m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90837e0efff74042a2382b302873e90b : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9d5eac5acc47f9e982b9d771ac4b1978196d50b6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-12 16:29:53.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/792642/4/check/tempest-dsvm-full-focal-py3/59b4e40/ : FAILURE in 1h 20m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/792642/4/check/devstack-plugin-ceph-dsvm-focal-py3/9d94299/ : FAILURE in 51m 19s","accounts_in_message":[],"_revision_number":4},{"id":"02f852fdf28785d0286567d679019ecfddb1dfb5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-12 16:36:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/42/792642/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1571630/1571630.html : SUCCESS in 2h 22m 38s","accounts_in_message":[],"_revision_number":4},{"id":"84335ab1afbda589cfa7a0ea8b535ed0ba81decf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-12 17:20:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/792642/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/90da819/ : TIMED_OUT in 3h 04m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/792642/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3648763/ : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":4},{"id":"9f6f97a686e5ae7c7ebb8e2613f169618618fdc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-12 17:38:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a7be009ddf1c8e5478a1f666436277fa039ee8a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 17:59:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27694557e6a64513a9f41d10e64b1e59 : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0969aa5f8f174a99b25c0a79cb3d53f4 : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4153c56accfbbc3e6cbb31ea563666aae6fab5c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-12 18:42:56.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/792642/5/check/iscsi-driver-primera/2952952/2952952.html : FAILURE in 1h 03m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ab983cbbca801fb3c5e4a323f3f8948fddc854cd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-12 18:54:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/792642/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3e3d9c9/ : SUCCESS in 1h 15m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/792642/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7dc00f1/ : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":5},{"id":"55aa41ac571d9876bdad5522c3a1e351365f6c0f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-12 18:57:03.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/792642/5/check/tempest-dsvm-full-focal-py3/6e010c2/ : FAILURE in 1h 14m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"aec4c69194b579fa09a0331b2126ee1c72ae83f5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-12 19:04:28.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/5/iscsi/ : FAILURE in 1h 25m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/5/smb/ : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":5},{"id":"2761df511cafb16f395375e03ad16646403dcfd1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-12 19:32:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-4/ns5_nfs : SUCCESS in 4h 34m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-4/ns5_iscsi : SUCCESS in 4h 40m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6864c0fdebe65a7924596c94bdd5a7191aebdcee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-12 19:47:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/42/792642/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0a28b5d/0a28b5d.html : SUCCESS in 2h 07m 53s","accounts_in_message":[],"_revision_number":5},{"id":"71c32cd47065d5f5676350d0e765b67ad5352496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 19:48:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45484d8ecb4c4d8bb90a2dcaf4533bf8 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/339b46a770f94b1fbeefd3fa7203decf : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41152a9ad6b4b1d975a50ee4f62c734 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e51d163b527149058e8ca86c8ade9264 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da27523ffd394a3cb5606e9d4257bcc8 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/db6d97847ee14310a0fbf14f5c26d490 : SUCCESS in 1h 01m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e306a749c284a099c8a721be0651dd2 : SUCCESS in 1h 11m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d523d568d2344e1aac175fd624273d27 : SUCCESS in 12m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2cd01baa07ed4838baee2884e5c79716 : SUCCESS in 6m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c36daf87b07849aeacf3e82cc9e20220 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cf9370193474ced9415089455bfebdf : SUCCESS in 6m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6f443191ef942b7ba016b78fc006d8b : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8c8b067e25a64bc483f9a54a9b45f220 : FAILURE in 32m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34197d4b95ca43e6aaccf00ce4cfdb35 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4023db44d6c34029a2a9613ae6b8150c : SUCCESS in 1h 17m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/20ef92a4b9bc4ee2a44276ad61aa7efe : SUCCESS in 1h 29m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f573a49cbd345c6a5dcbea129aeef01 : SUCCESS in 1h 39m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/289af1aa88074a7bb2e94f0ba3fcbf08 : SUCCESS in 1h 15m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8295492519c74abbbe9a0641ff4476a3 : SUCCESS in 1h 19m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/94d945002b0f475dada9708a9133a765 : SUCCESS in 1h 31m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72ea9731d4e6493e9c8085039c5ca162 : SUCCESS in 2h 08m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e7aaa6c874f47fa90204886904a1bb6 : SUCCESS in 49m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a3e9dd7a1500d7bf849bac4ea50c2f1b3009c83d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-12 20:46:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/42/792642/5/upstream-check/cinder-cDOT-iSCSI/e8ef72f/ : SUCCESS in 35m 33s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/42/792642/5/upstream-check/cinder-cDOT-NFS/f6dab4c/ : SUCCESS in 33m 29s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/42/792642/5/upstream-check/cinder-cDOT-FCP/03c934b/ : SUCCESS in 56m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"04929d0dfd1dfa44e56bb52370ce54fcfb830996","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-12 21:28:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/792642/5/ : SUCCESS in 1h 57m 13s","accounts_in_message":[],"_revision_number":5},{"id":"caec44875939f0651670abed100800ad4b263f54","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-12 22:37:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/42/792642/5/upstream-solidfire-check/cinder-solidfire-iSCSI/ec125e9/ : FAILURE in 2h 47m 40s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"5383ca27672dbea313b37da407b5146d81a24e03","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-12 23:33:38.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792642-5/PVME-iSCSI : FAILURE in 9:19.62\n* PVME-FC http://os-logs.tristero.net/dell-ci/792642-5/PVME-FC : SUCCESS in 47:12.25","accounts_in_message":[],"_revision_number":5},{"id":"064db65b81732ed92647752c414c4867e3e684e3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-12 23:36:52.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/42/792642/5/check/DellEMC_PowerStore_iSCSI/360e24a/DellEMC_PowerStore_iSCSI/239 : FAILURE in 1h 17m 14s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/42/792642/5/check/DellEMC_PowerStore_FC/744859d/DellEMC_PowerStore_FC/240 : FAILURE in 52m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"feb3a65fd701a0d690e51b2f6e9d0d83eec33977","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-13 00:09:54.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-5/ns5_nfs : SUCCESS in 4h 31m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-5/ns5_iscsi : SUCCESS in 4h 35m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"b8f46a8e7c2c7119947e03dca7dc445965d4b41e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-13 02:13:32.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792642-5/Seagate-iSCSI : SUCCESS in 41:03.22\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792642-5/Seagate-FC : SUCCESS in 50:14.88","accounts_in_message":[],"_revision_number":5},{"id":"7ff6fbd05ab5a4890ce6681787e635aab472f53e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-13 03:17:46.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792642-5/HPMSA-iSCSI : FAILURE in 1:01:02\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792642-5/HPMSA-FC : SUCCESS in 36:08.82","accounts_in_message":[],"_revision_number":5},{"id":"3b26f84059483c814345b31ef07df91c2313d5a8","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-13 03:40:32.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792642-5/Lenovo-iSCSI : SUCCESS in 51:13.66\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792642-5/Lenovo-FC : SUCCESS in 55:58.55","accounts_in_message":[],"_revision_number":5},{"id":"7ea1e7b0df0e7c03225c166c51df23bdb8ac49e7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-13 04:03:48.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/42/792642/5/check/DellEMC_XtremIO_FC/219e3f8/DellEMC_XtremIO_FC/490 : FAILURE in 1h 21m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/42/792642/5/check/DellEMC_XtremIO_iSCSI/6b6464e/DellEMC_XtremIO_iSCSI/480 : FAILURE in 2h 01m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"9f46f76fac5d5da7858f8632781d85bc4c04927b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-13 06:09:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/42/792642/5/check/DellEMC_Unity_FC/c442f55/DellEMC_Unity_FC/423 : FAILURE in 1h 24m 49s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/42/792642/5/check/DellEMC_Unity_iSCSI/2ad8534/DellEMC_Unity_iSCSI/398 : FAILURE in 1h 53m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"40c5d2fa69aec671c195326f7247d41a8a419e53","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-13 17:00:05.000000000","message":"Patch Set 5:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/792642/5/fc : SUCCESS in 0h 55m 09s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/792642/5/iscsi : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":5},{"id":"889463a14b69afd36957b05461f3300afa69b35e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-14 00:38:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/42/792642/5/check/DellEMC_PowerMAX_FC/7e5a911/DellEMC_PowerMAX_FC/250 : FAILURE in 6h 24m 24s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/42/792642/5/check/DellEMC_PowerMAX_iSCSI/e112bd7/DellEMC_PowerMAX_iSCSI/234 : FAILURE in 5h 20m 22s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"4806287ac75d5487cb3e5d17cf723b212a823884","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-15 12:41:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/5/20210715190445/NEC-FC : SUCCESS in 2h 32m 48s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/5/20210715190445/NEC-ISCSI : SUCCESS in 2h 37m 08s","accounts_in_message":[],"_revision_number":5},{"id":"570f755a0cf5a05a607cadf0ea035d4cd7f572ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-15 18:03:42.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"48712d34589bb161372dc7c630113daa4a7fa92f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:36:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1725e98100b74b479357de085d16f968 : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b5a0c0cb3aa416d85dbde422b667f5c : FAILURE in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a7bc864bc34af96e0d491bdbf254f292d0b2626","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 19:48:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/792642/6/check/iscsi-driver-primera/5964d2a/5964d2a.html : FAILURE in 1h 03m 28s","accounts_in_message":[],"_revision_number":6},{"id":"8d37dc97473af57cf87ef659fa4d737eb60fcf06","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-15 20:24:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/6/iscsi/ : SUCCESS in 49m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/6/smb/ : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7cecab7a994e65a9bd36488351e8dfc92152bfe4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-15 20:27:07.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"ad57e9a05843889d2fd6738f976861c128b5263b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 20:33:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b652452c5bb94bc4b7335d74a9af98e8 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bad4cb217a241a7bcfbaeae0a941137 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1e5ef0e3bc44150b0572a1c71211603 : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aaf8d80a120497e8b2fb824acd88ba2 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2e020d4ff154685a1975806458ec9d2 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/c5968307e1c245059b8ad9cee089e06e : SUCCESS in 1h 00m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e8d3d49804741d4bacbf7d298988c0b : SUCCESS in 1h 07m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4901f057f76945d888132dc4a296efe6 : SUCCESS in 12m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/28b2978d3fe844de9f3c2d54fa17a1b8 : SUCCESS in 6m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/866dcec9341048a09ed14b1fe4b99ebb : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c05dcfad18249938d7d0a8e05450cec : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/746fbd9c0256499683b80b8ac2796e24 : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/496f86cf86c7489a8a6529ea77d0c7c3 : FAILURE in 36m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc5e3b10a55f4dc996f73f9e21a1a672 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6d88d9f7f559439bb2ef9a895376a5ff : SUCCESS in 1h 26m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c74a420d4a4e465b8111a8547e763951 : SUCCESS in 1h 33m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13d052e091b94c7491f9cd10a4b5fae6 : SUCCESS in 1h 30m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6963e74e429242b2bc62b78bdf624eb5 : FAILURE in 46m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0273c6d192a4b23bb2208c7266f00ef : SUCCESS in 1h 21m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/acfe564e4c284cf3b2d7cdb13cad5f38 : SUCCESS in 1h 29m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28228708507c47da8bc7830cfefd81c2 : SUCCESS in 2h 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/690e8d681fe146619b34605ec7c5bbd0 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":6},{"id":"db96a4067d53eadc59a6e703919d1711dff79815","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-15 22:09:03.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/792642/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/30426dc/ : TIMED_OUT in 3h 02m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/792642/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/214279b/ : TIMED_OUT in 3h 02m 07s","accounts_in_message":[],"_revision_number":6},{"id":"0495a63e5a93c532ebb54a4ca6c772f1c11a308e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-15 22:09:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/792642/6/ : FAILURE in 43m 34s","accounts_in_message":[],"_revision_number":6},{"id":"ee973e94033451f9558aba681176be62532108f8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-15 22:14:23.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/42/792642/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8cadb5b/8cadb5b.html : SUCCESS in 2h 19m 38s","accounts_in_message":[],"_revision_number":6},{"id":"2574ef7fe902c1e1adbc5e3f468f4a65da482d12","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-15 22:43:20.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792642/792642/6/check-not-vote/ext-cinder-zuul/5a390b9 : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":6},{"id":"ba9cb46afec301e3a445824fd224c64925e705e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-15 23:34:37.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_Unity_FC/3b34f07/DellEMC_Unity_FC/472 : FAILURE in 1h 11m 48s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_Unity_iSCSI/0b3f4ab/DellEMC_Unity_iSCSI/443 : FAILURE in 1h 17m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"a635db70c1bf35863efe2a50eb7256e725a60088","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-16 01:02:44.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-6/ns5_nfs : SUCCESS in 4h 47m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-42-792642-6/ns5_iscsi : SUCCESS in 4h 54m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"8231bd4c9f42ad255b435777aeec6ac66149c7a6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 03:12:33.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_XtremIO_FC/ddae83f/DellEMC_XtremIO_FC/536 : FAILURE in 1h 19m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_XtremIO_iSCSI/41e176c/DellEMC_XtremIO_iSCSI/522 : FAILURE in 2h 39m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"28bc07013219fcb9e8284fe24c546f43fb2c0d3a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-16 09:35:40.000000000","message":"Patch Set 6:\n\nrecheck cinder-grenade-mn-sub-volbak - problem during setup, tests never ran","accounts_in_message":[],"_revision_number":6},{"id":"a96a4112a2004f60debb0099c2873383eda9558b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 09:56:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/663182f43f924cdf93e016849b93704c : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b69c97067944a738dc8d0b670d1b135 : SUCCESS in 19m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d736b472afb9981e3346ca93ea3b6d1c36bbcaa0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-16 10:07:36.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"a3c310308dc32024648ddfeb639b253bae5da7c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-16 10:10:46.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792642/792642/6/check-not-vote/ext-cinder-zuul/4de9282 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":6},{"id":"54e54deb49157a9bb8f935e0998dd1b0bfe1fd07","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-16 10:28:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/6/iscsi/ : SUCCESS in 52m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/6/smb/ : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":6},{"id":"84474d5e6977c4af1a92b239a26d4a594cf00a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 11:33:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3de1d3ce99a3407493eea99d77e7c0ee : SUCCESS in 19m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4e0cb90e8e44990972d94be01a4b277 : SUCCESS in 14m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a89d84918d47491f975de355b897f677 : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa00818c37ab4832b1a980320afae34c : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a07d4549db74443b6d7f38dc780336b : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/fe9812d9e8674742837f395cafa7bebc : SUCCESS in 53m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56290017b7884b6693f0d4dd144ecd15 : SUCCESS in 1h 16m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/86e4f7faf4864e3896a1c8113548a41d : SUCCESS in 11m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce85c3a38a5c499b93bd55261008a13e : SUCCESS in 6m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06b08eb1261c4a2eae8fad69c197b36a : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/173a4296eda94f97866554c2f484bd77 : SUCCESS in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5769f1df7597486190f8e8c05d63d205 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c76e696199404408a60c18a1a874b017 : FAILURE in 34m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2918d3b0a19f4d73a7e8ec978f20052e : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6999c19c66e4bdc8d2a8bf7c6ccceaf : SUCCESS in 1h 11m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/99f71e6be49a4b7e911af5d4df3d5611 : SUCCESS in 1h 48m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/807fdf33d0ac458287727f394b9a9d36 : SUCCESS in 1h 19m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a143889b4398495b9915e359af135807 : SUCCESS in 1h 34m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f4456238802140338d4367b241e283a5 : SUCCESS in 1h 16m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b70451e7f414975bfaec48aa2b65137 : SUCCESS in 1h 09m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/679ef8749a274528b89331c2c5bff4c4 : SUCCESS in 1h 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a6f47a1bb0a432b97399d3710c69d5f : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":6},{"id":"149bd4615b089304351217aa2268c89e3232945d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-16 12:22:10.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerStore_iSCSI/893e448/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerStore_FC/eb88025/DellEMC_PowerStore_FC/252 : SUCCESS in 47m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"00b6ba5cb7066fa5d5eafe4ad2754da20e15b869","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-16 12:35:57.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/42/792642/6/upstream-check/cinder-cDOT-iSCSI/1795005/ : FAILURE in 56m 43s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/42/792642/6/upstream-check/cinder-cDOT-NFS/e2963d9/ : SUCCESS in 37m 53s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/42/792642/6/upstream-check/cinder-cDOT-FCP/f0822a8/ : SUCCESS in 1h 00m 56s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f48aa2cc0f154123c4d5db5f93db2e13e6199f96","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-16 12:45:02.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_SC_FC/30afed4/DellEMC_SC_FC/171 : SUCCESS in 45m 02s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_SC_iSCSI/9999df7/DellEMC_SC_iSCSI/167 : SUCCESS in 1h 10m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"d8698c2b9b12eb0b432d9b453f9c51d9b181545a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-16 13:19:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/42/792642/6/upstream-solidfire-check/cinder-solidfire-iSCSI/71a6cda/ : FAILURE in 1h 43m 51s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"8cf816525e7395818b69db2180c4176a422327fe","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-16 14:06:42.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792642-6/Seagate-iSCSI : SUCCESS in 47:23.28\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792642-6/Seagate-FC : SUCCESS in 42:47.03","accounts_in_message":[],"_revision_number":6},{"id":"24f3615d9e351d54a5305f719d8e9b6a08cf84bd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 14:19:46.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerFlex/2efced9/DellEMC_PowerFlex/294 : FAILURE in 1h 34m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"c7180565c5d4767c97dafb760b103a2f9d46de4d","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-16 14:31:53.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792642-6/HPMSA-iSCSI : FAILURE in 1:04:13\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792642-6/HPMSA-FC : SUCCESS in 34:17.10","accounts_in_message":[],"_revision_number":6},{"id":"682a3cab51d73b04ea740c5565b987249c12d3a6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-07-16 14:33:48.000000000","message":"Patch Set 6: Code-Review+1\n\nDefinitely needed, so that people don\u0027t forget about it (even if it\u0027s non-voting for now).","accounts_in_message":[],"_revision_number":6},{"id":"c35b63449c136589c58a838d04351227a4400ccc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-16 15:18:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_Unity_FC/afb0bca/DellEMC_Unity_FC/480 : FAILURE in 1h 15m 16s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_Unity_iSCSI/8cb0de9/DellEMC_Unity_iSCSI/451 : FAILURE in 1h 12m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"42a84be7a672bd2ce03a3daed652b3250dcd2b77","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-16 15:22:25.000000000","message":"Patch Set 6: Code-Review+2\n\nCI passed, it\u0027s non voting and we really need this +2","accounts_in_message":[],"_revision_number":6},{"id":"da6ffa21f61bb7ef110205333c1c6d12ccd36291","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-16 15:29:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/42/792642/6/check/dsvm-tempest-cisco-zonemanager-job/f999a9a/ : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":6},{"id":"120377f30278de7c6ec39a7503d99f465e8e7b71","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 16:34:52.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_VNX_iSCSI/38caf6d/DellEMC_VNX_iSCSI/262 : SUCCESS in 2h 54m 56s\n- DellEMC_VNX_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_VNX_FC/5e4289c/DellEMC_VNX_FC/280 : SUCCESS in 1h 59m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"c3353e3f0762bfabac00a4b0705589d1b2808cf8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 18:49:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_XtremIO_FC/207e47b/DellEMC_XtremIO_FC/543 : FAILURE in 1h 16m 53s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_XtremIO_iSCSI/43b1f14/DellEMC_XtremIO_iSCSI/529 : FAILURE in 2h 58m 26s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"36e1dad173747fc1cb59873011968f3631ed80c8","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-17 03:21:05.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792642-6/PVME-iSCSI : FAILURE in 2:59:36\n* PVME-FC http://os-logs.tristero.net/dell-ci/792642-6/PVME-FC : SUCCESS in 47:50.87","accounts_in_message":[],"_revision_number":6},{"id":"f20ef5b7dad6a808dca9102ffb78e2dacffb6e7c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-17 05:09:58.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792642-6/Lenovo-iSCSI : FAILURE in 15:47.25\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792642-6/Lenovo-FC : FAILURE in 16:02.34","accounts_in_message":[],"_revision_number":6},{"id":"17e00f40bc6e086a74e66910ec99ffd2cf03d144","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-17 17:44:23.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerMAX_FC/6474350/DellEMC_PowerMAX_FC/266 : FAILURE in 5h 31m 18s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerMAX_iSCSI/59e3832/DellEMC_PowerMAX_iSCSI/252 : SUCCESS in 5h 29m 39s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"14c29c56574eb456e5c6a454ab135549407dc3a1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-18 17:34:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/6/20210718235435/NEC-ISCSI : SUCCESS in 2h 21m 31s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/792642/6/20210718235435/NEC-FC : SUCCESS in 2h 39m 22s","accounts_in_message":[],"_revision_number":6},{"id":"de4e6103dcb3101bbf9dceac389df63b83ddc50f","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-07-19 02:19:10.000000000","message":"Patch Set 6:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":6},{"id":"63446b57f8eb9879d8519ac08e6bd985a9a52b5b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-19 02:52:39.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792642/792642/6/check-not-vote/ext-cinder-zuul/c669ef6 : SUCCESS in 32m 45s","accounts_in_message":[],"_revision_number":6},{"id":"2b3f861cdc3cd9fb16c74ca38fcf1fe5f7038c0e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-28 16:00:39.000000000","message":"Patch Set 6:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/792642/6/fc : SUCCESS in 1h 10m 58s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/792642/6/iscsi : SUCCESS in 1h 15m 31s","accounts_in_message":[],"_revision_number":6},{"id":"2e3c0bbd3015f7fa6e3a406ecaa80f0351ea4afc","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-08-02 01:07:56.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"09cb9b427e1df2410b4a8ec4d9d10a00be4a9553","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-03 14:01:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nGood as non-voting.","accounts_in_message":[],"_revision_number":6},{"id":"7473a1810a5e51f894c1ae75f9eccaa2c0279ba2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 14:02:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4c0ad613dc2b09e8bdb55fbd8338571f469e2ce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 17:20:09.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6170729eaa1417bac390123eedabc34 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d4cc70e48a54c03ab0aeb656021921d : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f47af93e3841a4b73f9f7ef94568f4 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e349652aee42c5ae148b5d1d308a7d : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a2998f4060f94488b94d40bf42f60bbc : FAILURE in 58m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d1cd05c5fcd4b1b9edc55f2d0984267 : SUCCESS in 1h 09m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2f87a4df7d464c85bfaf07aab1547e1b : FAILURE in 1h 03m 01s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/08598f26c4fa47a3b7e7479fb52c50d1 : SUCCESS in 1h 14m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c6fba0f15ae4beaa844d06cf94ecad3 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":6},{"id":"91ee3a2b5ff416a8b125914c59cf3e007f7a069a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-04 13:15:43.000000000","message":"Patch Set 6:\n\ngrenade failure is addressed by https://review.opendev.org/c/openstack/grenade/+/803317, which merged after the tests were run on this patch","accounts_in_message":[],"_revision_number":6},{"id":"725602d2a88b17a0c6dfec01cf4ffd375c56a742","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-04 13:15:56.000000000","message":"Patch Set 6:\n\nrecheck grenade","accounts_in_message":[],"_revision_number":6},{"id":"1c2a11ebab948d094ce796e027ac364b58840616","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 13:38:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1649ba5851284b55949081fe95932fb5 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b03ac0fbeb314769b50a86dcbd88ca33 : SUCCESS in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9c2eda1fd452d0669f2210d3f6538838a23d9723","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-04 14:17:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/792642/6/iscsi/ : SUCCESS in 1h 00m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/792642/6/smb/ : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":6},{"id":"19b98bf2baf416a4fb59f8e410ef9ff5127dde78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:18:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1529a87a25408c90a741663d7204ed : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae71d88487444948929bcf19edbef219 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b23125b839e49789e94698931b4971b : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e2c4ad4047445f39a8e4e65274317b9 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc619163168d41eb9ce0eb8937bdfcf1 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/87192bfee66c4cc895e9a8d8b02407f7 : SUCCESS in 1h 09m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdc3ede189c048cc9754f038be42bc88 : SUCCESS in 57m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ddf7b3e3b9fd40b09e97d5a16fe17c15 : SUCCESS in 13m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7498f14824b44008ab640e281f815ca5 : SUCCESS in 5m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f4c5a5e44fd4564be397fb53cc7f83b : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb7cf55590264cdab9a58982208a9f61 : SUCCESS in 7m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/555aefd1e2a8494c95d34ab915355ac9 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b67f5a98c0747b1bb5eb04c7a4d5406 : SUCCESS in 44m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3d8466886c24321ab616ba0f1c31296 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a124fae4e9e649eab3119777fde9307b : SUCCESS in 1h 22m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b43087e08f084f9787763604487f25a5 : SUCCESS in 1h 28m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d4d069bceb54b529e1be4737f357af5 : SUCCESS in 1h 26m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bee9764c92724d0c80ae54aeaabef9a3 : SUCCESS in 1h 19m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4008fc9154c84e339abed19c4554c07c : SUCCESS in 56m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/36a95ee9e3df47b59c714a5d6b2f727c : SUCCESS in 1h 10m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6b7a63754a142f5a11c1ae6553df0d1 : SUCCESS in 2h 00m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29eb9bd33f6f46c7abcab266ec966272 : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":6},{"id":"55e52add4ba813e650a061e395774f0a94f01222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:19:21.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"367096b2a241b472ebffddeed518985667479ac3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-04 15:31:34.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/792642/6/check/tempest-dsvm-full-focal-py3/47025f4/ : FAILURE in 2h 12m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/792642/6/check/devstack-plugin-ceph-dsvm-focal-py3/04713bb/ : FAILURE in 48m 22s","accounts_in_message":[],"_revision_number":6},{"id":"8af1c6638575b963dcb2e290cb654b1ad38aac3b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-04 16:26:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/42/792642/6/upstream-check/cinder-cDOT-iSCSI/8a4f325/ : SUCCESS in 35m 07s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/42/792642/6/upstream-check/cinder-cDOT-NFS/efeeebe/ : SUCCESS in 35m 46s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/42/792642/6/upstream-check/cinder-cDOT-FCP/e22d21a/ : SUCCESS in 1h 07m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f1b1fb06a3b04b46b84e9b61353ae59f88bbf22d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-04 17:00:23.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/42/792642/6/upstream-solidfire-check/cinder-solidfire-iSCSI/fd1ba13/ : FAILURE in 1h 40m 54s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"699fbbc57befddcb0b55bd7d9ee9b90a63b41ff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:25:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc24782ec12347cfbb3a98f08b1cbf2d : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e6528b9164f45208444a20754a8e1ae : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bb0d7f8e3a34a4483cc8cb363865446 : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d5f3ebf1a084e749776fb5cfe918c4e : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/b4b42e04422341779c98ae3f03b45e44 : SUCCESS in 1h 06m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/010bf13696104eca8373b0c14c89bab6 : SUCCESS in 1h 12m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6b5f747e5e6494bab00f47467ccbf39 : SUCCESS in 1h 13m 14s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc2104ef51034845ae7714de3abb1200 : SUCCESS in 1h 09m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/def6cb3a575741c4a04041c4ca96d2a4 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":6},{"id":"dd0bf14ab002e295e4ddc9dda747577c4d1d1203","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:25:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7bda0c903f531fdb77d5764c602697e5bcb04f5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:29:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488c7b9bf71d42aaae000f62faba8bbe : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":6},{"id":"1664208983f5d378b593f2192e23bdfec7377b63","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-04 18:37:18.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerStore_iSCSI/7cc6c0b/DellEMC_PowerStore_iSCSI/294 : SUCCESS in 50m 48s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerStore_FC/6bf4655/DellEMC_PowerStore_FC/318 : SUCCESS in 48m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"3425e3da54d11cccd15a97611452e5b60af820b5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-04 18:56:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/42/792642/6/check/dsvm-tempest-cisco-zonemanager-job/923876f/ : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":6},{"id":"6ca03487ba60bc6886a567c8dc6b9d51407ece78","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-04 23:49:49.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_SC_FC/6da2859/DellEMC_SC_FC/235 : SUCCESS in 47m 33s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_SC_iSCSI/4c44de5/DellEMC_SC_iSCSI/231 : SUCCESS in 1h 16m 21s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"0d87f819fd6b194bd9da3e5f22a6c0757cb18c74","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-05 01:54:55.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/792642/792642/6/check-not-vote/ext-cinder-zuul/99e99b4 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":6},{"id":"2290854c3cb76ca10d5a4c67c74829bce34bddde","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-05 03:52:00.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/42/792642/6/check/DellEMC_PowerFlex/5c5145e/DellEMC_PowerFlex/352 : SUCCESS in 1h 39m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"087baba01aef63210ef7aefc8afd61335b405988","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-05 04:25:57.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_Unity_FC/8d0e3a0/DellEMC_Unity_FC/575 : FAILURE in 1h 18m 09s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_Unity_iSCSI/59ca2ff/DellEMC_Unity_iSCSI/545 : FAILURE in 1h 21m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"bc50655d555bb3f8f4322acfb7f8fcb6183bce9d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-05 06:47:32.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/792642-6/Seagate-iSCSI : SUCCESS in 40:16.75\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/792642-6/Seagate-FC : SUCCESS in 37:33.58","accounts_in_message":[],"_revision_number":6},{"id":"6c6393601b1c95ef79f372c515a8450ead1eaf7d","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-05 09:17:40.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/792642-6/HPMSA-iSCSI : FAILURE in 59:19.03\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/792642-6/HPMSA-FC : SUCCESS in 33:07.64","accounts_in_message":[],"_revision_number":6},{"id":"fe6a1c7d7b2096952e2bfcd307888d3c1647494f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-05 17:53:05.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/42/792642/6/check/DellEMC_XtremIO_FC/168eecf/DellEMC_XtremIO_FC/602 : SUCCESS in 1h 24m 49s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/42/792642/6/check/DellEMC_XtremIO_iSCSI/e8c372a/DellEMC_XtremIO_iSCSI/588 : FAILURE in 3h 01m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"feead368c8124451730621fd74fd366b7baf77f5","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-05 18:31:14.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/792642-6/Lenovo-iSCSI : SUCCESS in 53:51.02\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/792642-6/Lenovo-FC : SUCCESS in 50:36.61","accounts_in_message":[],"_revision_number":6},{"id":"50d10ba86421706c8e946fa72646b4e06cf736bb","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-05 20:23:45.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/792642-6/PVME-iSCSI : FAILURE in 1:35:21\n* PVME-FC http://os-logs.tristero.net/dell-ci/792642-6/PVME-FC : SUCCESS in 46:12.01","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cc6b499d23229c94f9a9bc6b272ec93dd94b6cba","revisions":{"3a2549032ef7d0f8423ae4ad62275d3553af9963":{"kind":"REWORK","_number":1,"created":"2021-05-21 14:46:28.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/792642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/792642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/792642/1"}}},"commit":{"parents":[{"commit":"ba27068fc0cc8a70e2a2e2fcfe6ab9896dab1908","subject":"Merge \"PowerMax Driver - Fix for deleting replication group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba27068fc0cc8a70e2a2e2fcfe6ab9896dab1908"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"subject":"Run cinder-mypy job in the check","message":"Run cinder-mypy job in the check\n\nPromote the mypy job from experimental to a nonvoting check job.\n\nChange-Id: I7aa9d603f69ba4424da7ae6f98192a888da1071c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a2549032ef7d0f8423ae4ad62275d3553af9963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a2549032ef7d0f8423ae4ad62275d3553af9963"}]},"branch":"refs/heads/master"},"94859ca58ff2384f25f3b33ca4ae55ff792d9d28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-09 17:20:55.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/792642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/792642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/792642/2"}}},"commit":{"parents":[{"commit":"1b5aad85d04b262281b9632bbc9a3ab490ee9082","subject":"Merge \"Abort volume creation when encryption spec is invalid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b5aad85d04b262281b9632bbc9a3ab490ee9082"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-09 17:20:55.000000000","tz":0},"subject":"Run cinder-mypy job in the check","message":"Run cinder-mypy job in the check\n\nPromote the mypy job from experimental to a nonvoting check job.\n\nChange-Id: I7aa9d603f69ba4424da7ae6f98192a888da1071c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94859ca58ff2384f25f3b33ca4ae55ff792d9d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94859ca58ff2384f25f3b33ca4ae55ff792d9d28"}]},"branch":"refs/heads/master","description":"Rebase"},"658e5ab76e69af62c179356651b8ad3921a4ed5a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-09 20:05:37.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/792642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/792642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/792642/3"}}},"commit":{"parents":[{"commit":"a2c3c6c846122071913f7c55289076ba67924a1d","subject":"Add installation of mypy stubs packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2c3c6c846122071913f7c55289076ba67924a1d"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-09 20:04:59.000000000","tz":-240},"subject":"Run cinder-mypy job in the check","message":"Run cinder-mypy job in the check\n\nPromote the mypy job from experimental to a nonvoting check job.\n\nChange-Id: I7aa9d603f69ba4424da7ae6f98192a888da1071c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/658e5ab76e69af62c179356651b8ad3921a4ed5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/658e5ab76e69af62c179356651b8ad3921a4ed5a"}]},"branch":"refs/heads/master"},"0a71b5cb91924d9b6c789d9713d43519c09e2148":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-12 14:11:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/792642/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/792642/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/792642/4"}}},"commit":{"parents":[{"commit":"6d68de4825ebd0e46c2942b6c4d0995771fba97b","subject":"Add installation of mypy stubs packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d68de4825ebd0e46c2942b6c4d0995771fba97b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-12 14:08:45.000000000","tz":-240},"subject":"Run cinder-mypy job in the check","message":"Run cinder-mypy job in the check\n\nPromote the mypy job from experimental to a nonvoting check job.\n\nChange-Id: I7aa9d603f69ba4424da7ae6f98192a888da1071c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a71b5cb91924d9b6c789d9713d43519c09e2148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a71b5cb91924d9b6c789d9713d43519c09e2148"}]},"branch":"refs/heads/master"},"52fa450c1810b890fd589e790c0ee7fb0cf34b84":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-12 17:38:07.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/792642/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/792642/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/792642/5"}}},"commit":{"parents":[{"commit":"a51eda5f08aa5e6780bc192013ad53f95288b6db","subject":"Add installation of mypy stubs packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a51eda5f08aa5e6780bc192013ad53f95288b6db"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-12 17:36:38.000000000","tz":-240},"subject":"Run cinder-mypy job in the check","message":"Run cinder-mypy job in the check\n\nPromote the mypy job from experimental to a nonvoting check job.\n\nChange-Id: I7aa9d603f69ba4424da7ae6f98192a888da1071c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52fa450c1810b890fd589e790c0ee7fb0cf34b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52fa450c1810b890fd589e790c0ee7fb0cf34b84"}]},"branch":"refs/heads/master"},"cc6b499d23229c94f9a9bc6b272ec93dd94b6cba":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-15 18:03:42.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/792642/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/792642/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/792642/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/792642/6"}}},"commit":{"parents":[{"commit":"c8c4fc953c4a563ed2295efaf82c75b52de06a2c","subject":"Add installation of mypy stubs packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8c4fc953c4a563ed2295efaf82c75b52de06a2c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-05-21 14:45:52.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-15 18:03:42.000000000","tz":0},"subject":"Run cinder-mypy job in the check","message":"Run cinder-mypy job in the check\n\nPromote the mypy job from experimental to a nonvoting check job.\n\nChange-Id: I7aa9d603f69ba4424da7ae6f98192a888da1071c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc6b499d23229c94f9a9bc6b272ec93dd94b6cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc6b499d23229c94f9a9bc6b272ec93dd94b6cba"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
