)]}'
{"id":"openstack%2Fcinder~793169","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I52b87d03c49db74002f09cfeea86bc0ac7ce8a42","project":"openstack/cinder","branch":"stable/ussuri","hashtags":[],"change_id":"I52b87d03c49db74002f09cfeea86bc0ac7ce8a42","subject":"Open local image files with \"rb\" mode","status":"MERGED","created":"2021-05-26 14:30:28.000000000","updated":"2021-06-25 02:33:22.000000000","submitted":"2021-06-25 02:31:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793169","meta_rev_id":"3e0b7b71fa8d71ca7f2fb43b602bddde43501c7c","_number":793169,"virtual_id_number":793169,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 02:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-06-18 17:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-06-24 10:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2021-06-18 15:29:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-23 13:26:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2021-06-18 14:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":2,"date":"2021-06-24 12:23:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":1,"date":"2021-06-24 12:23:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-26 14:36:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-26 15:43:13.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-26 15:47:06.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-05-26 16:25:32.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-26 17:19:17.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-26 17:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-26 20:47:52.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 17:44:12.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-06-18 14:39:06.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-06-18 15:29:35.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-06-18 17:28:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-23 13:26:51.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2021-06-24 10:03:04.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-06-24 12:23:43.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"e5e8c282a2e6a96f240e430c14b55dd8b9e0a556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-26 14:30:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a9aeb3430a66cb15384eeda097ce7bc05ec1174","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-26 14:36:09.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7ee0fb2c7974fc07ab96ccf1ee1a5a88cd18ff6a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-26 14:38:18.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7a824cdd46f00ae0ac898bb4873a57450d5a7716","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-26 14:58:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/69/793169/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1d09ec7/1d09ec7.html : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":1},{"id":"09e164d3fae25fa856a47f2742a16652f02a35f7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-26 15:43:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793169/1/iscsi/ : SUCCESS in 1h 04m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/793169/1/smb/ : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cf07d94765a52ed0a1a4b0a533209898cf607db5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-26 16:25:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/69/793169/1/check/dsvm-tempest-cisco-zonemanager-job/0c63711/ : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e1e9885f932171e64369c646bdc7e4cd38aecb05","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-26 17:19:17.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/937ba7b4cfe041aaadb5e380c7283dee : POST_FAILURE in 1h 55m 22s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/5b6e0cd7456644e789dbf26937e6367b : POST_FAILURE in 2h 08m 33s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/a13cdef65faf430da815b391d226ce6f : POST_FAILURE in 1h 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c209a2b4644578892376df914a24fea09a224a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 17:33:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a018ff5973824225b92473605c4488ff : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e84bdf2dcdb34f58b42248192c51f977 : SUCCESS in 9m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0dfd0056d954e2ab94492c9d8b5ab20 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6893bcc8616642469820283ae9d4e533 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/59c43f5183d34703991a86efa6eb4f51 : SUCCESS in 1h 03m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a6134d7091547f4af60e6a69b9cf9ad : SUCCESS in 1h 06m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/268da497a9f14c078ba791fd6208d119 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e7171bf2aa9457c86dc7172c7c1042d : SUCCESS in 7m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4f6fa4c6d8504f8aa6d43548d0ea4694 : SUCCESS in 9m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fc6df4f20c56457bbc1b3fbcb1df5d02 : SUCCESS in 33m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05e83b133410448fa3fa7f85e77be4c0 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72a37a7dbd9e426d8f05517235996c72 : SUCCESS in 1h 15m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7df3096ff42e422cb43a1e56e9cd4db8 : POST_FAILURE in 2h 46m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8170ae9494a5417fa9574858971a0194 : SUCCESS in 1h 01m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a943bc866b4c425986c280e2f244b978 : FAILURE in 2h 36m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eca21d74960e4e559c65c75a5e9a6056 : FAILURE in 1h 14m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38afc20a2199406db46f9d41afcddad3 : FAILURE in 2h 27m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6172c3f8b4d143aeb7451c2786ccd7c0 : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":1},{"id":"efc80bd40379c2ec85bc5553622207f8de04acb6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-26 20:47:52.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-69-793169-1/ns5_nfs : SUCCESS in 4h 56m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-69-793169-1/ns5_iscsi : SUCCESS in 4h 31m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"fcd72bf215d8bc9257f4583c48ac627b90f4c84a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-05-27 13:24:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fad3de460ba1502936d24b9fad3cf01b22d0d5a5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-27 14:14:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/69/793169/1/check/dsvm-tempest-cisco-zonemanager-job/5d64691/ : FAILURE in 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"97a9ed8ab3da4419a63ef0cad51ad06ff58bf1e7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-27 14:35:19.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793169/1/iscsi/ : FAILURE in 1h 09m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/793169/1/smb/ : SUCCESS in 1h 07m 35s","accounts_in_message":[],"_revision_number":1},{"id":"941e259233830fc24570d4b3950ed79d25d01c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:49:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0247f7c96cca498498e3c0e314928f06 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7b244fb57e646caa2181f73d39c1339 : SUCCESS in 10m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7159eddccbff4e3e8d2e8fc2a3e833cc : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7950bd6e75724edc8dbe08999f07ce24 : SUCCESS in 17m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/c2f706ef360a4f808a5d07cebc899449 : SUCCESS in 1h 21m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d43dc4b11cb406387c43576ef2307df : SUCCESS in 1h 16m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c5fdca8aba242be9054b863645a03b5 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e2bb820376c488aa9509c1f3fc7e413 : SUCCESS in 7m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/579330118dcc42e0bf309c6362fe4ed7 : SUCCESS in 7m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e35696f0bc2046ef93bb571eda5064a0 : SUCCESS in 45m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42b33bf829c7449595621478d19ad38a : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a01f979d99e541b8b78baeddbcabecbb : SUCCESS in 1h 14m 18s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/34664958a8fa4202ad1a4e5f8c03a6b5 : SUCCESS in 1h 51m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d8c1937f2454b23bdca672f24ee0ebc : SUCCESS in 1h 28m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/80a316f95a954e3da59de326b13ac14e : SUCCESS in 1h 31m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ffe9f27a1b3644d49849ecada798a440 : FAILURE in 1h 14m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b45e681318d43f7b19d6776c4023090 : SUCCESS in 2h 18m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273fc665586f46d090a341e955758d8a : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b2c59bb640fb2fd7bbd519eec59100e850da6673","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-27 17:06:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/fe8ab1fa057443a08bc6535e499c5fda : POST_FAILURE in 1h 50m 29s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/6c081e02704247cd983d8376073c9c4e : POST_FAILURE in 1h 50m 41s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/6e88cc3b62254277aaafa8a4582b0826 : POST_FAILURE in 1h 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"97edac804fda1f987f0c4ea80b7b2d0d9b922879","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-27 17:44:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/69/793169/1/upstream-solidfire-check/cinder-solidfire-iSCSI/852f63c/ : FAILURE in 55m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"72d3a8d9699fe798a4e5821eabec6409a886b7fc","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-06-18 14:39:06.000000000","message":"Patch Set 1: Code-Review+1\n\nlooks like a clean backport","accounts_in_message":[],"_revision_number":1},{"id":"a95915d682a32433cfbb1c2006de1f9e54af2eb3","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-18 15:29:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e14e4ec5f70f9518d56846b359cdbc874c8cf33c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-18 17:28:21.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is needed in all the python 3 only branches.","accounts_in_message":[],"_revision_number":1},{"id":"fc7bfb78141594abb6d63c803071368be5036590","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-06-23 13:26:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b44cda2349cf4fa868192b93c7f7cd61717c5092","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-06-24 10:03:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c539832e323fa7169547c54cad114c788efedbc3","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-06-24 12:23:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood backport.  Ready to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b10cc67742619a356835752dad285b3d13617638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 12:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc10f9955e4d7f1c9ab4acdfaebbba9a6fc5ed9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 17:19:56.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a1ed6dde4f5decaf16d683f437bbeca6182bc19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:31:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285a3c77042d49f0ad9ba446180fe366 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99d0a364eccc46fa94c520e70c9a76aa : SUCCESS in 10m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51136d1f88634b1398f2c1d657966a1e : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b88b413eb464eefaf1c5c9d87c95307 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/71e90d8b66854d128e23da73c17f407c : SUCCESS in 1h 12m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24c906abe24e4fae9e456a25ceb01563 : SUCCESS in 1h 14m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13a89ce791fb4d79b851d7ef6a9e71f6 : SUCCESS in 1h 20m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35ff5d1cf5b34941ae6a857a87614836 : SUCCESS in 1h 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fbc0bff2ca34924b608d8466106ac29 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"974a6e6233cb48beb0fa24c193f99057ffb85419","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:31:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e0b7b71fa8d71ca7f2fb43b602bddde43501c7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 02:33:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2def96d5b40f42399c3a8313d83d1087 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6d7b09195ce738576252736708a7f3417f761db","revisions":{"f6d7b09195ce738576252736708a7f3417f761db":{"kind":"REWORK","_number":1,"created":"2021-05-26 14:30:28.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/69/793169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/793169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/793169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/793169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/793169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/793169/1"}}},"commit":{"parents":[{"commit":"e156653d39885b39026ecb08df315b7c0b0b8a55","subject":"Fix sporadic cleanup unit test failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e156653d39885b39026ecb08df315b7c0b0b8a55"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-04-12 13:24:47.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-26 14:30:17.000000000","tz":-240},"subject":"Open local image files with \"rb\" mode","message":"Open local image files with \"rb\" mode\n\nWhen using \"file\" in allowed_direct_url_schemes,\nopen image files w/ binary mode.\n\nCloses-Bug: #1923313\nChange-Id: I52b87d03c49db74002f09cfeea86bc0ac7ce8a42\n(cherry picked from commit 5d7e7547f0119300ecb8f6cd171b04ead3210ecc)\n(cherry picked from commit c16567bd8784761d24b73678377994903b8dc35b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6d7b09195ce738576252736708a7f3417f761db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6d7b09195ce738576252736708a7f3417f761db"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
