)]}'
{"id":"openstack%2Fcinder~793264","triplet_id":"openstack%2Fcinder~master~Idb14185d6d2559a9674d0abf5c6114d16d853dd2","project":"openstack/cinder","branch":"master","topic":"bug/1930267","hashtags":[],"change_id":"Idb14185d6d2559a9674d0abf5c6114d16d853dd2","subject":"Add timeout when execute ssh cmd.","status":"NEW","created":"2021-05-27 02:44:12.000000000","updated":"2021-09-19 01:12:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7cf3493b00530ada702da0db265f358bbbaaab5f","_number":793264,"virtual_id_number":793264,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-22 03:23:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},{"value":-2,"date":"2021-09-19 01:12:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-19 01:12:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 02:51:33.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 02:52:01.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 03:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 03:16:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 03:29:00.000000000","updated_by":{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},"reviewer":{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},"state":"REVIEWER"},{"updated":"2021-05-27 03:44:04.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 03:49:07.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-05-27 03:59:15.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-05-27 04:53:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 07:19:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 23:35:46.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-08 20:30:33.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 14:18:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 16:41:06.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 16:53:36.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-06-09 17:09:54.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-06-09 18:03:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 18:33:38.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-09 18:41:23.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-06-09 18:54:03.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 19:40:16.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-18 14:24:29.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-06-18 14:30:14.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-06-22 01:06:42.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2021-06-22 04:11:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-06-22 14:47:05.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"2ec0633891be98875d2438e00ce99b59d7fb88c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-27 02:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1efd2d272aa7607cfc7e40c263378260b4e525a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-27 02:51:33.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8661f2c125f9772bf17c3d67ea0c19a3ec500b7e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-27 02:52:01.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"719a0414a475b46fe15d7e6ce89e97a53e0c9f54","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-27 02:52:02.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"930df487ec3369254e3888f286d407bd61edd918","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-27 02:52:03.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"10e42b76db8cb7bc0e08c3c03f19c2aadc22f47d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 03:04:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b442d81de75409aad15ffd18f5e89b7 : FAILURE in 12m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d85137fb5f87434f94b482e2dd1f5d8e : FAILURE in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3872b087649d89852d67b34ae05a6188077e5287","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-27 03:16:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":1},{"id":"ae5bfd3328edc5950a1cfa663687cc9f33343e4c","author":{"_account_id":23729,"name":"焦浩霖","display_name":"jiaohaolin","email":"jiaohaolin01@inspur.com","username":"inspur-storage"},"date":"2021-05-27 03:29:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ebf775674ca284f0796a8ce53614689817fe05b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-27 03:44:04.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/3581feca28e643e2aa79443fdb30c24b : POST_FAILURE in 37m 43s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/bfe4c7248a9646138eb10624e48ec2d2 : POST_FAILURE in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bf48b844ddfc7320a430aef53c3154562df144a4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-27 03:49:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/793264/1/check/dsvm-tempest-cisco-zonemanager-job/c167e4f/ : FAILURE in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"10c91017dbdc07993c818395b88793b8d9ead5bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-27 03:49:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"60d84ac4b834dff5b5a40a1bffc06bcc485edfda","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-27 03:59:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793264/1/iscsi/ : SUCCESS in 1h 14m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/793264/1/smb/ : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d41dc6ef5a0db6cfe1b3f3e80d00772f61f76c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 04:53:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beee55945a4d4bbe8ed84478929290e5 : FAILURE in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db3116b8b10d41a1a9e8ac77fda1c1ac : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e674aea000ee4a4e86fdd352106394e0 : FAILURE in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76f0018493f64f7f8a9a0959b0eb3465 : FAILURE in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/410648be4b41434b82201cb54aa86f10 : FAILURE in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fedd8ef9ffaf495ebbaacef4ec833edf : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/82306177776b46abb0e2733f8ffb1c20 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c22bbecf90341e3b9295c147a646440 : FAILURE in 1h 45m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb557872046b400e8766d74260d3d3cb : FAILURE in 12m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06fd09a8935c4e609f43d08410671f95 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c83a7495b294b2d9b9723c8b8db241e : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/406c35622e7c4488b690cf84058e667c : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1fde8060584a4541954245d384295108 : FAILURE in 27m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/097e1ae41d4e400ea1477213721f9a54 : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/37be8dc180ff4039971d1db8a8075bbf : SUCCESS in 1h 31m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/74323198ac6f4296a9686e692bca3a48 : SUCCESS in 1h 31m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cd9579a43e96435f970b72d1114c3db0 : SUCCESS in 1h 28m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/755357a5b6f24b4382daf476ad6fd280 : FAILURE in 1h 12m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5e809de41d20429f8cb805284ffa7797 : SUCCESS in 1h 21m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f8d84134e97647df91b152e74e70d841 : SUCCESS in 1h 24m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a28149e8bdc4c5689652b0dfa1f3c14 : SUCCESS in 2h 08m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78112e6fe1144a7994f00692689033a8 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"54c3feb702172650d30adcf700673300f416015a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-27 07:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-1/ns5_nfs : SUCCESS in 4h 31m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-1/ns5_iscsi : SUCCESS in 4h 34m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"0e077b313fe6e26c50d877c5a151a6c06e85b387","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-27 07:28:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cfc9cc5f91f237307032d8beab7ec04e6d872137","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-05-27 23:35:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/793264/1/20210528040244/NEC-ISCSI : SUCCESS in 2h 35m 04s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/793264/1/20210528040244/NEC-FC : SUCCESS in 4h 32m 59s","accounts_in_message":[],"_revision_number":1},{"id":"50c03b47cdb0228f9d43df20c54184721ef8e0df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-31 12:01:06.000000000","message":"Topic set to bub/1930267","accounts_in_message":[],"_revision_number":1},{"id":"b2345dc9a68615a44c998eadce9437abb8209766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-31 12:01:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"77f81a549e830742039fbce33e9df534a542fb00","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-31 12:07:19.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c7a603a2efc3ca003968dd515585ae7a14d8e556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:22:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/799a3769184248f0b3eeb15b8537c481 : FAILURE in 19m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60b43ad5e3f94688bdb86acae11657fe : FAILURE in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"391e0706277d12058bc924e3f626be61b1cd336d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-31 12:51:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/793264/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e66c8a8/e66c8a8.html : FAILURE in 25m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c7d459aff034332ea3529992e9aa3b3d70e9b16f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-31 13:02:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/793264/2/check/iscsi-driver-primera/db0f70c/db0f70c.html : FAILURE in 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"09566d1521f3506f70f529e71a9f71ddfbe1be9a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-05-31 13:23:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793264/2/iscsi/ : SUCCESS in 59m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/793264/2/smb/ : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2249f8cabd807533578534e20845694d784220a2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-31 13:50:08.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/2/check/tempest-dsvm-full-focal-py3/c178041/ : SUCCESS in 1h 42m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/2/check/devstack-plugin-ceph-dsvm-focal-py3/965f2bd/ : FAILURE in 41m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6204dd67902197ac55e7af97b2a6eb70e36fc210","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-31 13:55:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"76ee4c1c9e123ec2b7934b605030e7f376150e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:16:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5254afcb9884effb5860784dcb9502d : FAILURE in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f47262eb49040709ad0ae3a7f7a745b : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc0b202613494592b1f339fb52651730 : FAILURE in 10m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80011e3b4a6845c09056e58ef319ddbb : FAILURE in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a85a63147264d90ac62a4749d1998d5 : FAILURE in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a5a72b87da454b9d6744566cd702c6 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8a35936e0b8241f7a72cbb9c02bb7a6d : SUCCESS in 1h 02m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a71556a8c799445e8ec4279e13c54765 : SUCCESS in 1h 27m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2cf200215d344f2b8d9afd4fe2719d30 : FAILURE in 12m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e94c6e5413c4f219887eadba9cc6382 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/690a0e2e90c240bf832715c2aee95e1b : SUCCESS in 6m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39d9a7a3ff194060a59992c2c4984605 : SUCCESS in 8m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebd831fa1e8448e4bbc90123bd7583b9 : FAILURE in 32m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c80a9460df64473b91f42db06dcfa62 : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e581f074cc434ff6a1ac40249d854f06 : SUCCESS in 1h 16m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/09f6d9ecb4f841f6918f30c3bdef9b41 : SUCCESS in 1h 38m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/026b90eda0464aa09a27159bf666a100 : SUCCESS in 1h 42m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3be39d348cc94934829bd0a2ee0e5b26 : SUCCESS in 1h 10m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/62e5affe70604ea6becb4b479520365a : SUCCESS in 1h 30m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/44c1b48b9fcb40a6800d0eeeae442cd5 : SUCCESS in 1h 33m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/631f463df09b4a3fa5e81e22c2027827 : SUCCESS in 2h 11m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c1f3481ff2a4253ac67204fa21fe198 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":2},{"id":"24a7f1d675409bbf0f6c048b6a8e9c0b032b8cc8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-31 14:53:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : MERGER_FAILURE in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"60d8ae408c72d671807ba8d69d6677a7238bac91","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-31 15:16:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/793264/2/check/dsvm-tempest-cisco-zonemanager-job/e871a8c/ : FAILURE in 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0627d877f4a5d004efb727e5bb6cec3bbf4af4c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-05-31 15:31:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":2},{"id":"973f8800dc9bca06c22cfa60e3a43e3da52f70aa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-01 00:46:04.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-2/ns5_nfs : SUCCESS in 5h 10m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-2/ns5_iscsi : SUCCESS in 5h 5m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"17b5fc137ca8ee1d4dd4ade1e1badd77d027d4b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-01 00:54:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"428678455b1f2b5e099eb9080284a8bf0d5d7ca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-01 02:57:03.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c1d8b5be8274ffd19bf2b1012e994765337b0718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-01 03:03:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"13eb12f122d931aea554ed09058303e11cedcaf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-01 03:05:40.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/64/793264/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"28f04b113256807ff01ad70dfebe91a62512ea7c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-01 03:07:37.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"a3f69701f9246ccfec3f0248c9a01c81b8d93a49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 03:24:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be306a81cedd4276b9f1a1c98a87d26e : SUCCESS in 16m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fb1db73246d42c4ab28630a4820e42d : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"11d685ca453545215f274f063bf422e9794b7827","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-01 03:30:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/793264/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bdc2acc/bdc2acc.html : FAILURE in 26m 03s","accounts_in_message":[],"_revision_number":4},{"id":"d1186f324c8d0c41ef3b945ec09f5787939c5153","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-01 03:35:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/64/793264/4/check/iscsi-driver-nimble/c7b8c39/c7b8c39.html : FAILURE in 27m 20s","accounts_in_message":[],"_revision_number":4},{"id":"7c1a5897033f88a0ca66028fe7464715f62ea909","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-01 04:03:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":4},{"id":"81c950df01ff5ad143f3d36fbf571f2b4f50f091","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-01 04:10:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/793264/4/check/dsvm-tempest-cisco-zonemanager-job/dcb9e0b/ : FAILURE in 1h 01m 26s","accounts_in_message":[],"_revision_number":4},{"id":"6ffeeabe342e51e0289037a2ea9660a83332c451","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-01 04:11:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793264/4/iscsi/ : SUCCESS in 1h 07m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/793264/4/smb/ : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8f96a1f2e7e4a329fee4dfe7b1b37f41e928df05","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-01 05:00:28.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/4/check/tempest-dsvm-full-focal-py3/8795c7d/ : SUCCESS in 1h 43m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/4/check/devstack-plugin-ceph-dsvm-focal-py3/68b22d6/ : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":4},{"id":"aee52499d255bbc13226afc49f96ca771eb0af2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 05:07:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/504105ff242d4e10b44a9325dd5f63ec : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4acbbf0bfe84001a380cc99bf724ad1 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec8903d60518411899b36c69c7e2d527 : SUCCESS in 14m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e41a489c4434766b98e8b445a20d0ec : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bdac50b82394b06b344a44094acaa10 : SUCCESS in 16m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d562a513aed429e96fe0b7a14f4810a : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/68256d8261ae42b79c19489bf01f6d07 : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfd4cbc5cf0e42e4b3516c7483bad479 : SUCCESS in 1h 05m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc75a28fbd124c86ac47c3368f6e3341 : SUCCESS in 13m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b765a73c47c24677b3ac76ae258b8c3c : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b923a9a4675442abd615ab2070ea874 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af16f47fe01444b7aec2551abdb5f060 : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aa523c98177341d5b8c976ac06ad8309 : FAILURE in 30m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/970a5f2263904518ac1faa3f37606216 : FAILURE in 9m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f7cd328c10e540e2b0e937138b508a1c : SUCCESS in 1h 14m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7c1dd8234864497a9b30b4bdea4aef5a : SUCCESS in 1h 31m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f47b4087671c4199842a53b845552759 : FAILURE in 1h 22m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e33c22452140462cae234cf7fb2a0a28 : SUCCESS in 1h 10m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/04d1478733dd4af18648a7447c238347 : SUCCESS in 1h 22m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a16e5d93f8884292acd8c33f4f3d8691 : SUCCESS in 1h 17m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/145b8f8d81454869873bc8b28c0e8943 : SUCCESS in 2h 03m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef1840dd94504ae7a6bdccc4ffe1b524 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f00972c53316bc6c0c9ba1b3a569df018a406b06","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-01 07:25:46.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-3/ns5_nfs : SUCCESS in 4h 25m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-3/ns5_iscsi : SUCCESS in 4h 28m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"432b6ea3dc360dbd8c1adaf48cb97ae781593972","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-01 08:25:38.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-4/ns5_nfs : SUCCESS in 4h 42m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-4/ns5_iscsi : SUCCESS in 4h 35m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"9740b4bfc9a8cb7b542696216a110ee295467a69","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-01 15:32:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/793264/4/20210601215532/NEC-FC : SUCCESS in 2h 35m 38s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/793264/4/20210601215532/NEC-ISCSI : SUCCESS in 2h 36m 32s","accounts_in_message":[],"_revision_number":4},{"id":"06d58ffec30865e03e2d366aeefd065be0a9a76a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-08 20:30:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/64/793264/4/ : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":4},{"id":"828dd1e80db9cdaf40d596b31227680ab01c345d","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-09 13:38:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c9f494f96bbeba3e6464bcbe17924109e98dcc2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:03:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e138e25504c94e0b93000b51decf528b : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa1ee4eb0f764fd9b1788c3910e8d711 : SUCCESS in 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"91aaa602b45ab31799387d5db0f27d0180f03d6d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-09 14:11:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":4},{"id":"4587f6a8275d9de46f5c46615a93400c5f3105d2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-09 14:18:36.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/793264/793264/4/check-not-vote/ext-cinder-zuul/2926b00 : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":4},{"id":"a922b6a4b211d37da8cd5383538fb10c9454bfe4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-09 14:43:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/64/793264/4/check/dsvm-tempest-cisco-zonemanager-job/a4eaace/ : FAILURE in 58m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8978c3b52d9c664e34c98ced725c755bfc02f8e6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-09 14:49:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793264/4/iscsi/ : SUCCESS in 1h 10m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/793264/4/smb/ : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":4},{"id":"63a5419737cdda5a2ce6a6ba81aff3145b352a83","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-09 15:31:52.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/4/check/tempest-dsvm-full-focal-py3/f088f84/ : SUCCESS in 1h 43m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/4/check/devstack-plugin-ceph-dsvm-focal-py3/3d69772/ : FAILURE in 52m 48s","accounts_in_message":[],"_revision_number":4},{"id":"51d007487176175f0d7fd65c05e4c12bf157d03b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:42:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4aef21a5c854e25bb2d5a0492702a09 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768fd52fd53c4f72866d701cf1400488 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40cea3aad2c84f48824dd67530e19bf9 : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a9c9f8f660c4e2390a94786bc52a226 : SUCCESS in 11m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08000b8718214967bba65cfcd3ea2416 : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/932957b78b974a2cb55b25ea0573c489 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/c8e3174866f4472186e7b9cff4ae3f39 : SUCCESS in 1h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7bf02931d5cc47ee954ddfd0f92706f6 : SUCCESS in 1h 09m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c12fcbe7fcdf41df9d08e0ccbcb2d554 : SUCCESS in 14m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7963f8e43144098930a7ac5c1be5780 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6c12746690c45358c4a75c53bf83fd6 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98d3fbd020f645e8b3dcf383b76389c4 : SUCCESS in 8m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/234934be281f4a878436bf09e7e872f4 : FAILURE in 43m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e58bb836179445439dacf2c850cf25dd : FAILURE in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/558c9a85db9b482d810c9910ddbccbfa : SUCCESS in 1h 40m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2e0ab7b766b24758a0bcd785724a9755 : SUCCESS in 1h 46m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eace55e57dbd450fac841379be0a564f : SUCCESS in 1h 42m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f62974d6aeeb428ea3823f5514d29a13 : SUCCESS in 1h 18m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b287d83557248bfb99d12a162816ef3 : SUCCESS in 1h 28m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/792eaeb5d8924d9c96b80f832a7f33e5 : SUCCESS in 1h 25m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ec3dc85fa799405a924cb04bac7b5239 : SUCCESS in 1h 59m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/360a414ad1574eaaae62d9376b94800b : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":4},{"id":"43032f718414bea7560f18d2644bd7f91ff87d53","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-09 16:41:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/64/793264/4/upstream-check/cinder-cDOT-iSCSI/d6a2f8e/ : SUCCESS in 37m 21s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/64/793264/4/upstream-check/cinder-cDOT-NFS/a7a164c/ : SUCCESS in 39m 37s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/64/793264/4/upstream-check/cinder-cDOT-FCP/ff54205/ : SUCCESS in 55m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"e87cca33658eaab36e844d413514d2a56e44abe5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-09 16:53:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/64/793264/4/upstream-solidfire-check/cinder-solidfire-iSCSI/241f78b/ : SUCCESS in 1h 08m 11s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"1a5cf5478b03d100ce2f6d8adc84ba473351805e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-09 17:09:54.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/64/793264/4/check/DellEMC_PowerStore_iSCSI/5e40a79/DellEMC_PowerStore_iSCSI/137 : SUCCESS in 1h 24m 47s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/64/793264/4/check/DellEMC_PowerStore_FC/bdf2faa/DellEMC_PowerStore_FC/138 : SUCCESS in 1h 22m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"cd88101440149e01d32e849468bd0e8e4c60a4a5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 18:03:20.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/64/793264/4/check/DellEMC_PowerFlex/f1f9304/DellEMC_PowerFlex/158 : SUCCESS in 2h 18m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5564f315359e29a1bbadc9a613ca3bac6786e357","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-09 18:33:38.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/64/793264/4/check/DellEMC_Unity_FC/ed375b2/DellEMC_Unity_FC/266 : SUCCESS in 1h 53m 42s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/64/793264/4/check/DellEMC_Unity_iSCSI/d66d09e/DellEMC_Unity_iSCSI/250 : SUCCESS in 2h 12m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"71a91fe906931e78369b22f93ff277d26bdd8cfb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-06-09 18:41:23.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/64/793264/4/check/DellEMC_SC_FC/6a654bf/DellEMC_SC_FC/157 : FAILURE in 1h 25m 57s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/64/793264/4/check/DellEMC_SC_iSCSI/d437d12/DellEMC_SC_iSCSI/154 : FAILURE in 2h 56m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e6a3509379fca8dff9aa68e9006e45896341c4c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 18:54:03.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/64/793264/4/check/DellEMC_XtremIO_FC/7b2f404/DellEMC_XtremIO_FC/256 : FAILURE in 1h 54m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/64/793264/4/check/DellEMC_XtremIO_iSCSI/81e7a16/DellEMC_XtremIO_iSCSI/259 : SUCCESS in 2h 22m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"01a2c02940911f0db62963b4d4602b0100031b83","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-09 19:40:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/64/793264/4/check/DellEMC_VNX_iSCSI/24b4e31/DellEMC_VNX_iSCSI/152 : FAILURE in 3h 55m 35s\n- DellEMC_VNX_FC http://publiclogs.emc.com/64/793264/4/check/DellEMC_VNX_FC/4739971/DellEMC_VNX_FC/157 : SUCCESS in 1h 57m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"e3185c61bf6f4c6f10e1a7009312ccb43b9bfd87","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-06-18 14:24:29.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks good","accounts_in_message":[],"_revision_number":4},{"id":"db97905b5fd5b6e5bc8a364f923ffebd6706ad3a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-18 14:30:14.000000000","message":"Patch Set 4: Code-Review-2\n\nI think this change is not correct.\n\nThe option is described as \"connection timeout\" but this change makes it so that ssh commands across various drivers cannot run longer than 30 seconds even if they are actively performing an operation.  The timeout parameter passed to oslo_concurrency\u0027s ssh_execute is a limitation on waiting for the command to finish, not a TCP timeout.\n\nHowever, in other drivers that use paramiko instead of oslo_concurrency\u0027s ssh_execute, this option is applied as a TCP timeout rather than an operation timeout, which is significantly different.","accounts_in_message":[],"_revision_number":4},{"id":"afe4179c045277852920e1330ad4878c78e07dd6","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-21 03:02:35.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-2\n\u003e \n\u003e I think this change is not correct.\n\u003e \n\u003e The option is described as \"connection timeout\" but this change makes it so that ssh commands across various drivers cannot run longer than 30 seconds even if they are actively performing an operation.  The timeout parameter passed to oslo_concurrency\u0027s ssh_execute is a limitation on waiting for the command to finish, not a TCP timeout.\n\u003e \n\u003e However, in other drivers that use paramiko instead of oslo_concurrency\u0027s ssh_execute, this option is applied as a TCP timeout rather than an operation timeout, which is significantly different.\n\nThanks for review. \nNormally, the execution will not time out. However, in the high concurrency scenario, when the storage pressure is relatively high, it may occur. It is necessary to provide a configuration of execution timeout.\n\nCan I add a new configuration item to solve this problem?","accounts_in_message":[],"_revision_number":4},{"id":"78df893f59e62fe5707c575dca0f353408d0f0ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-06-22 01:06:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"103603e2cdc910c81e4c6f15884dd93a61da3a20","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2021-06-22 01:06:42.000000000","message":"Patch Set 5:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":5},{"id":"c831405bb0d7a5a4f5a5e1ffc6d3a29e8fbd1d86","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-22 01:07:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"c36e91d723d0cb06b2a0b4b5693dffdedf9d15bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 01:27:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70672dcb50ea4846a5e57f406adf42a2 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bc50a7a90e34684a40f315f2bb07384 : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e708b6bf3f3d0eeab8bd92186c8727e068c13e47","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-22 01:42:00.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/793264/793264/5/check-not-vote/ext-cinder-zuul/d1b9a79 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":5},{"id":"b46a614a13efbe99daef2fac5f605f6292dc74e0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-22 01:43:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/64/793264/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9b6ea2b/9b6ea2b.html : FAILURE in 23m 53s","accounts_in_message":[],"_revision_number":5},{"id":"973702abd94a2c4813a98a8f91e18e7f48565e5f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-22 01:48:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/64/793264/5/check/iscsi-driver-primera/fd76d82/fd76d82.html : FAILURE in 29m 46s","accounts_in_message":[],"_revision_number":5},{"id":"b7f785e01322c64db57d8cf66cd22d256ffb9d96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-22 02:01:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"7e2f19d2d28722b56855183f58153abb50d82560","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-22 02:12:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/793264/5/iscsi/ : SUCCESS in 54m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/793264/5/smb/ : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":5},{"id":"7b71105796ab89c9c22fcec861455ef64fd04dca","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-22 03:05:31.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/5/check/tempest-dsvm-full-focal-py3/a09f9ce/ : SUCCESS in 1h 49m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/64/793264/5/check/devstack-plugin-ceph-dsvm-focal-py3/cc527b9/ : FAILURE in 45m 37s","accounts_in_message":[],"_revision_number":5},{"id":"ed2aa0823488717b717d83163431499e65d30ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:23:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdba56a019df49e49ebc991fb074ddf4 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b099aac5eb4fc291720693a600d0d0 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eefc6fa897464ab2a6cff79768ef3940 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8b421f95a914bbe9af7d14eced1cf55 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73de6006fdb7444988299e5718ddac2f : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4327c2c7f0242a38e2053bb7edb7fcb : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e9c28573c09e44dc9cf6cc925af99e47 : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/734aa1a775cc40f7a82ff1f01e388c91 : TIMED_OUT in 2h 07m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2db6e8f0836147ea890824ff21816d01 : SUCCESS in 22m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a892fdb204814f1ba749b66fcc6e2a19 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d1a06c15fe04a61b8b88a223f25a912 : SUCCESS in 9m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/47e9a8e7980841a7b94e31c8e0c1e7f2 : SUCCESS in 10m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f01cd931d2514748b54a4b5af95a87c5 : FAILURE in 30m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/871c7d3526d94821b804c22eeffa06b3 : FAILURE in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/38f6be04788a4076b83879ffa8c3ea5b : TIMED_OUT in 2h 05m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b6a8009618134449895d37f6ca36bb06 : SUCCESS in 1h 19m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/50500d9d00f941e08f1bc349cc37f69e : SUCCESS in 1h 19m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6b98befd6ed64ab2815de64b3e258237 : SUCCESS in 1h 10m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/81f5cfbca8de449c8f1bceacb999537e : SUCCESS in 1h 14m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/35331bd6a4cb4ddcbf736a1ff88ed9b3 : SUCCESS in 1h 07m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/786363121a204e8fa5d920c8c80ab9d6 : SUCCESS in 1h 59m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b4f37faee6404c8d89957f3647a207 : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":5},{"id":"119a6b19327e0eab5b786a6285eee16b16b09e24","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-22 04:11:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/793264/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d412b49/ : TIMED_OUT in 3h 03m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/793264/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/46cd9e0/ : TIMED_OUT in 3h 03m 13s","accounts_in_message":[],"_revision_number":5},{"id":"43f6ec93f7d7520d918877507a18b2260a2ee489","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-22 07:12:54.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-5/ns5_nfs : SUCCESS in 4h 28m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-793264-5/ns5_iscsi : SUCCESS in 4h 33m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"24e948f7cdd82c149c96e3f7e0b449024cd56576","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 08:51:39.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/64/793264/5/check/DellEMC_XtremIO_FC/dadf28b/DellEMC_XtremIO_FC/363 : SUCCESS in 1h 21m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/64/793264/5/check/DellEMC_XtremIO_iSCSI/7c99f16/DellEMC_XtremIO_iSCSI/364 : SUCCESS in 1h 50m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"3799709a0fdc2a454cc7352a55adfe20cb978a81","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 14:47:05.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/64/793264/5/check/DellEMC_PowerMAX_FC/7258770/DellEMC_PowerMAX_FC/177 : FAILURE in 4h 17m 06s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/64/793264/5/check/DellEMC_PowerMAX_iSCSI/193691a/DellEMC_PowerMAX_iSCSI/163 : SUCCESS in 4h 00m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"2552f05609a5ee76d402b96a3648703dbea78fbc","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-23 18:13:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/793264/5/20210624003934/NEC-ISCSI : SUCCESS in 2h 32m 42s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/793264/5/20210624003934/NEC-FC : SUCCESS in 2h 34m 03s","accounts_in_message":[],"_revision_number":5},{"id":"4da884a2d01c8294a0a163abd8b20cade478aeb6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-09-19 01:12:33.000000000","message":"Topic bub/1930267 removed","accounts_in_message":[],"_revision_number":5},{"id":"3b64a77385f5d2bce2d57177099d1f5b9267dd7c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-09-19 01:12:38.000000000","message":"Topic set to bug/1930267","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7648c9b7d5968fe00d9895d4e8b3ffdf0d5e5cab","revisions":{"3ffe3ae441dfd385b62fededc276ccd9d4412a64":{"kind":"REWORK","_number":1,"created":"2021-05-27 02:44:12.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/64/793264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/793264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/793264/1"}}},"commit":{"parents":[{"commit":"da2fbff247f23c7b82e77db2c1e7772cb22801ac","subject":"Fix functional jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da2fbff247f23c7b82e77db2c1e7772cb22801ac"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-27 02:44:01.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-27 02:44:01.000000000","tz":480},"subject":"Add timeout when execute ssh cmd.","message":"Add timeout when execute ssh cmd.\n\nChange-Id: Idb14185d6d2559a9674d0abf5c6114d16d853dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ffe3ae441dfd385b62fededc276ccd9d4412a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ffe3ae441dfd385b62fededc276ccd9d4412a64"}]},"branch":"refs/heads/master"},"6f72775e07926bde433526a47541d66a2a273f4c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-31 12:01:38.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/64/793264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/793264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/793264/2"}}},"commit":{"parents":[{"commit":"da2fbff247f23c7b82e77db2c1e7772cb22801ac","subject":"Fix functional jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da2fbff247f23c7b82e77db2c1e7772cb22801ac"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-27 02:44:01.000000000","tz":480},"committer":{"name":"Xuan Yandong","email":"xuanyandong@inspur.com","date":"2021-05-31 12:01:38.000000000","tz":0},"subject":"Add timeout when execute ssh cmd.","message":"Add timeout when execute ssh cmd.\n\nCloses-bug: #1930267\n\nChange-Id: Idb14185d6d2559a9674d0abf5c6114d16d853dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f72775e07926bde433526a47541d66a2a273f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f72775e07926bde433526a47541d66a2a273f4c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c38835bd4b286dc5b6a8381e41d9baf7e1ac7df2":{"kind":"NO_CHANGE","_number":3,"created":"2021-06-01 02:57:03.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/64/793264/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/793264/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/793264/3"}}},"commit":{"parents":[{"commit":"da2fbff247f23c7b82e77db2c1e7772cb22801ac","subject":"Fix functional jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da2fbff247f23c7b82e77db2c1e7772cb22801ac"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-27 02:44:01.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-06-01 02:56:42.000000000","tz":480},"subject":"Add timeout when execute ssh cmd.","message":"Add timeout when execute ssh cmd.\n\nCloses-bug: #1930267\n\nChange-Id: Idb14185d6d2559a9674d0abf5c6114d16d853dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c38835bd4b286dc5b6a8381e41d9baf7e1ac7df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c38835bd4b286dc5b6a8381e41d9baf7e1ac7df2"}]},"branch":"refs/heads/master"},"f048ab5dcc3a70c5d91872a322f453937d0a7a30":{"kind":"REWORK","_number":4,"created":"2021-06-01 03:03:21.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/64/793264/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/793264/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/793264/4"}}},"commit":{"parents":[{"commit":"da2fbff247f23c7b82e77db2c1e7772cb22801ac","subject":"Fix functional jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da2fbff247f23c7b82e77db2c1e7772cb22801ac"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-27 02:44:01.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-06-01 03:02:59.000000000","tz":480},"subject":"Add timeout when execute ssh cmd.","message":"Add timeout when execute ssh cmd.\n\nCloses-bug: #1930267\n\nChange-Id: Idb14185d6d2559a9674d0abf5c6114d16d853dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f048ab5dcc3a70c5d91872a322f453937d0a7a30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f048ab5dcc3a70c5d91872a322f453937d0a7a30"}]},"branch":"refs/heads/master"},"7648c9b7d5968fe00d9895d4e8b3ffdf0d5e5cab":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-22 01:06:10.000000000","uploader":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"ref":"refs/changes/64/793264/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/793264/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/793264/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/793264/5"}}},"commit":{"parents":[{"commit":"ff5df92863ec339ab4f144f50c25c0266119111f","subject":"Merge \"Unit test RBD clone depth calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff5df92863ec339ab4f144f50c25c0266119111f"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-27 02:44:01.000000000","tz":480},"committer":{"name":"jiang guodao","email":"jguodao@vmware.com","date":"2021-06-22 01:06:10.000000000","tz":0},"subject":"Add timeout when execute ssh cmd.","message":"Add timeout when execute ssh cmd.\n\nCloses-bug: #1930267\n\nChange-Id: Idb14185d6d2559a9674d0abf5c6114d16d853dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7648c9b7d5968fe00d9895d4e8b3ffdf0d5e5cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7648c9b7d5968fe00d9895d4e8b3ffdf0d5e5cab"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
