)]}'
{"id":"openstack%2Fcinder~795009","triplet_id":"openstack%2Fcinder~master~I179f23d3e6c62d02c069be0dd3d60968a2119568","project":"openstack/cinder","branch":"master","topic":"bug/1931049","hashtags":[],"change_id":"I179f23d3e6c62d02c069be0dd3d60968a2119568","subject":"Add libcgroup related packages in bindep.txt","status":"MERGED","created":"2021-06-07 01:15:35.000000000","updated":"2021-07-16 17:29:27.000000000","submitted":"2021-07-16 17:25:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"795009-bug/1931049","meta_rev_id":"c3c2fc83383f4dcbe5b2eb007f008e7a1d280a87","_number":795009,"virtual_id_number":795009,"owner":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-16 17:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-07-16 15:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-06-24 10:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2021-06-15 03:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2021-07-16 14:41:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-18 00:54:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2021-07-16 16:04:51.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2021-06-18 15:21:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2021-07-16 16:04:51.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-07-16 15:26:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 01:23:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 01:31:28.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 01:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 01:48:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 02:17:48.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-06-07 02:25:24.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-06-07 03:02:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-07 03:13:51.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 03:22:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 03:25:37.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 03:58:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 05:40:01.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 03:26:57.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-10 03:09:01.000000000","updated_by":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"reviewer":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"state":"CC"},{"updated":"2021-06-10 03:22:07.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-06-10 03:32:12.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-06-10 04:09:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-10 04:18:56.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-10 05:25:08.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-06-10 06:13:10.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-11 23:20:21.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-06-13 02:24:22.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-06-14 00:20:24.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-06-14 00:20:24.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-06-14 13:30:22.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-06-14 13:30:22.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-15 03:04:02.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-06-15 04:24:00.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-06-18 00:54:49.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"CC"},{"updated":"2021-06-18 00:54:59.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-06-18 15:21:01.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-06-24 10:02:55.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-07-16 14:41:15.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-07-16 16:04:51.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"092ac7fac1d4602f81436aaa1443b09081ce24c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-07 01:15:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f57cf70d7b6daad32f0d6ff9bb3bba3759891c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-07 01:23:24.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"db0ff0cad6a023f4155c8e918ac7f93eb1a19109","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-07 01:31:28.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/795009/795009/1/check-not-vote/ext-cinder-zuul/754a71f : FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"729b8e9ea58105b9a42ca11464ba824bec631431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 01:33:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cb16a313b93418aa509b5056507af46 : RETRY_LIMIT in 2m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a717532fe4e949b18e7e0cc8f23e12d6 : RETRY_LIMIT in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a98585c0bf6add162da2b419ac0061051d8ed80a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-07 01:44:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/09/795009/1/check/iscsi-driver-primera/1e30d91/1e30d91.html : FAILURE in 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3841b8d290f4d1383e8d930c61945107d18dfcb0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-07 01:48:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":1},{"id":"ec4e878edad71b8c5be1ebae907c7e04e321dc39","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-07 02:17:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/1/check/dsvm-tempest-cisco-zonemanager-job/015d5eb/ : FAILURE in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a5bdca44237a1b85da467a4dd8d6a95b558154ae","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-07 02:25:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/1/iscsi/ : SUCCESS in 1h 08m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/1/smb/ : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"68703208e3414adc817e96c5c17b8aa6c4859754","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-07 03:02:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/09/795009/1/check/DellEMC_Unity_FC/7b2746c/DellEMC_Unity_FC/241 : SUCCESS in 1h 23m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/09/795009/1/check/DellEMC_Unity_iSCSI/4ebf83d/DellEMC_Unity_iSCSI/227 : SUCCESS in 1h 46m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ceae2ff19585d7220b405e41637481c8e1ca951d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 03:13:51.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/09/795009/1/check/DellEMC_XtremIO_FC/3918c47/DellEMC_XtremIO_FC/230 : SUCCESS in 1h 17m 07s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/09/795009/1/check/DellEMC_XtremIO_iSCSI/c3b7126/DellEMC_XtremIO_iSCSI/235 : SUCCESS in 1h 57m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"37498412ade34ff03cdd986c0b468a64ff273003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 03:22:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/56aeb37405b04d159061f57fd441bbb4 : RETRY_LIMIT in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64eee7a78fd54dd59552080688df46d2 : RETRY_LIMIT in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c287d56b404d069ffa86ff6192fc7d : RETRY_LIMIT in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/348b9d8378fa4553b96d05f460b9fcf3 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42fbd4f714274428a9e81f765a3c32a2 : RETRY_LIMIT in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15a0de4f17884d9db5ed60280bb3b092 : RETRY_LIMIT in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef000f469a8f44aeb1f57f14c441870c : RETRY_LIMIT in 2m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/d35b9bf8a3aa4a8aba78c3febeaa6dcc : SUCCESS in 59m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7ddf61199f349fb893093c26af67ea3 : SUCCESS in 1h 06m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a46197e07743e8a5dff5395723c20e : RETRY_LIMIT in 3m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0cbd4bb46e1f4b6c860ec8b190102add : RETRY_LIMIT in 2m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c966feaef38549ba83de029e05db918a : RETRY_LIMIT in 2m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6984430fac04e6ca0f5fb993d8e206f : SUCCESS in 7m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a681836664244d9952afac231b541a9 : RETRY_LIMIT in 2m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ae612fcc67a4a8d8be2451ab4f81046 : FAILURE in 29m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/953caaed6b594d848f7d7e670e4e585b : RETRY_LIMIT in 2m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ec0aee8b1314e8fad3e418b7735caef : SUCCESS in 1h 15m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0374491c9656429b9500ed94ec5f8964 : SUCCESS in 1h 30m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c07cf1935c4e4ba8bb137d2e60734678 : SUCCESS in 1h 29m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a174338d4cc9435095aa4b680f75e18f : SUCCESS in 1h 08m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/02bb4311806e430f9ba699cef8389889 : FAILURE in 1h 46m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/46aaaf8807a3416eae1bf9ef002bb81f : SUCCESS in 1h 17m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2dab7692de5742018d2770db6f206900 : SUCCESS in 2h 04m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efa4f9fadb4e4fe580afea69292d4575 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e5a5af710b249e2d43024d710becae548152fd7d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-07 03:25:37.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/09/795009/1/check/tempest-dsvm-full-focal-py3/2e507d7/ : SUCCESS in 2h 05m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/09/795009/1/check/devstack-plugin-ceph-dsvm-focal-py3/136ceaf/ : FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ebe7e2fdb314e6c9aad315e4968acab65d155829","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 03:58:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/795009/1/20210607102146/NEC-ISCSI : SUCCESS in 2h 26m 42s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/795009/1/20210607102146/NEC-FC : SUCCESS in 2h 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6222b9f2e913ed274aec0ac82ccdabab132e79ce","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-07 05:40:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-1/ns5_nfs : SUCCESS in 4h 21m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-1/ns5_iscsi : SUCCESS in 4h 23m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3013743cee1cc1f600a45a5757379f6357089343","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-08 00:40:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c97b547a082bc86ccb7ba9fb7aeeef2324fc3a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-08 00:45:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32f9c72bed89c4201efa2e3b5c165219f2c8f181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-08 00:45:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c0920d95368ae1a0fc2a676c4de34e6bda7a4078","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-08 00:49:36.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"1b205e3b721acf2548755e323736f8a6bd02ad26","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-08 00:51:47.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"a46ee616e263d6eb47e29960e2c1d7bb2b8946ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 01:15:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20b21651f6314bd6bd992680eba8850d : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e3c9f09791b494da613b83215686b14 : SUCCESS in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e03889fc4be7b033973f916bd13807aa27c44a10","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-08 01:47:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":3},{"id":"bb1d0a95b314a75e15317e8f677510300d225ee8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-08 01:48:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/3/iscsi/ : SUCCESS in 52m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/3/smb/ : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":3},{"id":"2d1aff597fa29948bc63dc6768f73845d39a1149","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-08 01:49:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/3/check/dsvm-tempest-cisco-zonemanager-job/d2e644b/ : FAILURE in 58m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9a0a7146464d26cf5f191abe3ec9656f5f433b2e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-08 02:30:46.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/09/795009/3/check/tempest-dsvm-full-focal-py3/154116d/ : SUCCESS in 1h 39m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/09/795009/3/check/devstack-plugin-ceph-dsvm-focal-py3/24bca79/ : FAILURE in 42m 10s","accounts_in_message":[],"_revision_number":3},{"id":"fc54780dbe6087910f2fec466a0fcfe6cbfb6d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 03:02:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3e2c0eeab65f49ff8c6092494cb6d586 : SUCCESS in 12m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f9d74863a7f441cad4716cc30bfa53d : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40943f865e8545bda2ec5eb7a6c5de12 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61a20607e4c4db9946fef103923e1f5 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9919e068de4e416ba8deaf42b7e2b3bf : SUCCESS in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ceee2bd661b4034a477dbf4d730c87b : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3e62ed238a43b3a5b1970913d5bba6 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/6cfbe2bd34a441b8b830b39e03b90334 : SUCCESS in 1h 09m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68b4aaf9831b4665b0359008f5e60c6f : SUCCESS in 1h 18m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f53d31d2e5f645abbc6a0c5c59f36199 : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c1c41b13834048618e99a2b273b2fcc5 : SUCCESS in 15m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4cf266b84b444cc926534c01b654d4d : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80cf929ac12b447cbf34873066abe394 : SUCCESS in 10m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/340c0d65150e4346933b144b04a7ef12 : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0370f5cf5b9b4efa81bbe90405e003fd : FAILURE in 36m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fab0bf05b8be4fd99f06892ff6fd0343 : SUCCESS in 7m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e92c90523c8649dc9c1886b83e6b0f2c : SUCCESS in 1h 30m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a91683206a0c4b96973cde4d86d1df42 : SUCCESS in 1h 35m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e0dda24b0c9b49cc93f7c1a55baca21a : FAILURE in 1h 32m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cbcb353922654f0fb6bc27aa62a81cb3 : SUCCESS in 1h 12m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dfaeb4fffe654c50b35940f8330d27df : SUCCESS in 1h 26m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a12a08a96a142808885d6f3b8334202 : SUCCESS in 1h 22m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4bd5c714da4148f2bc5a889525166b5a : SUCCESS in 2h 15m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04ed015561934fc3ae92af71965ff021 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9aa2d8772ed11fafd2b36e63b4040a58f1efd70f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-08 04:15:19.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_Unity_FC/1f17c84/DellEMC_Unity_FC/256 : FAILURE in 6m 29s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_Unity_iSCSI/728673c/DellEMC_Unity_iSCSI/240 : SUCCESS in 1h 46m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"25a1a2fe93f952bdd6988495fff39a49580ce506","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-06-08 04:41:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/795009/3/20210608110127/NEC-ISCSI : SUCCESS in 2h 31m 58s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/795009/3/20210608110127/NEC-FC : SUCCESS in 2h 39m 41s","accounts_in_message":[],"_revision_number":3},{"id":"417c795061045af8b58818f17f1d1c71bced1730","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-08 05:14:14.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-2/ns5_nfs : SUCCESS in 4h 23m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-2/ns5_iscsi : SUCCESS in 4h 28m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5dcce24cd62135abd55eee1553c13793eca4d390","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-08 05:46:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_XtremIO_FC/931886e/DellEMC_XtremIO_FC/245 : SUCCESS in 1h 12m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_XtremIO_iSCSI/65fc66b/DellEMC_XtremIO_iSCSI/248 : SUCCESS in 1h 46m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"f30cc5f193b3f8e4d1c1b5a596181a4b6597557c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-08 06:05:03.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-3/ns5_nfs : SUCCESS in 4h 28m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-3/ns5_iscsi : SUCCESS in 4h 33m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"683a48d9bfda735bf2e682b588914fd1152817ef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-08 06:46:18.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/795009/795009/3/check-not-vote/ext-cinder-zuul/693a9a5 : FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":3},{"id":"3a5de27c6ab584f26335f89159eb03c56652fd7b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-09 03:26:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/795009/3/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":3},{"id":"a0e4e618e64da22a0f4fe65432b60c9765dbcbe9","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-10 00:15:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1ff3ab7f56e54b12c25a727163e1ade89740e0e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 00:37:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fffc0ca9ed734638b14079aef8a873cb : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f811a719c6de4881b3fb2d088f531c15 : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a5c396c055b77a1f1c4ac418914922ad093f3fe5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-06-10 00:47:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":3},{"id":"334b0c4c67b9d4afec3644c9cf540aaf341a13bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-10 00:50:27.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/795009/795009/3/check-not-vote/ext-cinder-zuul/bf396fd : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":3},{"id":"d44eddd032614a7428120e0716600f2159200215","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-10 01:18:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/3/check/dsvm-tempest-cisco-zonemanager-job/e2cae27/ : FAILURE in 57m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ad38064b0d8fb5ace7c64b21f74c7f96608727ef","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-10 01:21:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/3/iscsi/ : SUCCESS in 1h 04m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/3/smb/ : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8e83c886d481576935b1e10d1e191b40598e9a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 02:27:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad14f59cb0e845deb8a1d0215d5c2b73 : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e1f70b0d60a4dfba564338bb4f7a4bf : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc3df4ab3c34eab8368e65ecd74aa00 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/116592fe3c194a9e9539719b507ea24b : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/948697e7e29847dd8ab996004214831c : SUCCESS in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1be025bc31c24bea99e21656457c5194 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7755602fb88e428185f70cb49abe0846 : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/edd8aa099b3d441b917510bbe71643e8 : SUCCESS in 1h 11m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5046ea14874f4b33ac7dcbf372d37780 : SUCCESS in 1h 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b08b0ebde704efb8d059521369ae8ba : SUCCESS in 6m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95ab3ea4e3b4453bb7877bb1df6225e0 : SUCCESS in 13m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e13dc25d111d47a4abd19358e7fc5d92 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/687bf64f9f06442eaf5ad126514c4025 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dad3c573cf164f8b911aad3cc43a1595 : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fa649b2be7654250ac428f3445a2fcee : FAILURE in 37m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0389f32f02c94640a3dd60714171afaa : SUCCESS in 7m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9789acedca974dcbb36ebd1ef7b9d705 : SUCCESS in 1h 22m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0976ee93bcfe45919cac45f7d8ccedd1 : SUCCESS in 1h 35m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/be68cb553085419896df034dc478402a : SUCCESS in 1h 31m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/36059b3327134c62a3f904823bf56b72 : SUCCESS in 1h 17m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b242004ac0d40eb9a7fe404ffa6255e : SUCCESS in 1h 18m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6481708f58e1466c9012729af95346f4 : SUCCESS in 1h 21m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15d1c8bcdf674e11936a892d47736691 : SUCCESS in 2h 06m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66636169a504fbf959e9b228c6a8e0c : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":3},{"id":"59b0a66b2436dfc2e10193e860e6b3dc2091b3c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-10 02:37:15.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/09/795009/3/check/tempest-dsvm-full-focal-py3/0234e2a/ : SUCCESS in 1h 49m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/09/795009/3/check/devstack-plugin-ceph-dsvm-focal-py3/9f68e0b/ : FAILURE in 41m 16s","accounts_in_message":[],"_revision_number":3},{"id":"f58f2ec8073c30f4034cb545262d4fc051839a21","author":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"date":"2021-06-10 03:09:01.000000000","message":"Patch Set 3:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":3},{"id":"56cbbf7b0fc5360344798c4a07627907c8ba6eb3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-10 03:22:07.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_PowerStore_iSCSI/fcacede/DellEMC_PowerStore_iSCSI/140 : SUCCESS in 54m 19s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_PowerStore_FC/9d938eb/DellEMC_PowerStore_FC/141 : SUCCESS in 52m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"01f4fce042811c14de3ec71a37e511ed560c952d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-10 03:32:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/09/795009/3/upstream-solidfire-check/cinder-solidfire-iSCSI/21dfaf1/ : SUCCESS in 1h 03m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"525e12628991527749fc3c894d035fb90e295d89","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-10 03:51:54.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/795009/795009/3/check-not-vote/ext-cinder-zuul/8574d73 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":3},{"id":"db5c68b8639f707aab2d70010d24d2e3855a3140","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-10 04:09:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/09/795009/3/check/DellEMC_PowerFlex/7a09e4a/DellEMC_PowerFlex/161 : SUCCESS in 1h 41m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"98206d7d91c9335a25de6ef21874e2050adef566","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-10 04:13:47.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_Unity_FC/32325c7/DellEMC_Unity_FC/271 : SUCCESS in 1h 39m 03s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_Unity_iSCSI/cda2421/DellEMC_Unity_iSCSI/255 : SUCCESS in 1h 46m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e25651c3a46d8b292803fbb15262790bb5385e21","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-10 04:18:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/09/795009/3/upstream-check/cinder-cDOT-iSCSI/905aac6/ : FAILURE in 1h 50m 33s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/09/795009/3/upstream-check/cinder-cDOT-NFS/b960f92/ : SUCCESS in 1h 39m 24s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/09/795009/3/upstream-check/cinder-cDOT-FCP/638314a/ : SUCCESS in 58m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"95ddfd9f2bb878c22e034211c1d682ea6a68c1ff","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-10 05:01:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_XtremIO_FC/1222852/DellEMC_XtremIO_FC/261 : FAILURE in 1h 17m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_XtremIO_iSCSI/ddf728f/DellEMC_XtremIO_iSCSI/264 : SUCCESS in 2h 01m 43s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"de379821204c454638862b99c48fe75235797e3f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-06-10 05:25:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_SC_FC/cc99abe/DellEMC_SC_FC/160 : FAILURE in 56m 06s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_SC_iSCSI/cdc32d2/DellEMC_SC_iSCSI/157 : FAILURE in 2h 57m 31s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"296423c9181e7dc3eb3acfbc968068677d9177e9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-10 06:13:10.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/09/795009/3/check/DellEMC_VNX_iSCSI/fe6c9b3/DellEMC_VNX_iSCSI/155 : SUCCESS in 3h 45m 00s\n- DellEMC_VNX_FC http://publiclogs.emc.com/09/795009/3/check/DellEMC_VNX_FC/e4f1fe5/DellEMC_VNX_FC/160 : SUCCESS in 1h 22m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"c81a77737310d0ccfb4339704d88921b9dc4478d","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-11 23:20:21.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/795009-3/HPMSA-iSCSI : FAILURE in 1:02:25\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/795009-3/HPMSA-FC : FAILURE in 23:00.86","accounts_in_message":[],"_revision_number":3},{"id":"5cc2e3c8cd01a7d3a248d949d8f869249db19d10","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-13 02:24:22.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/795009-3/PVME-iSCSI : FAILURE in 39:10.18\n* PVME-FC http://os-logs.tristero.net/dell-ci/795009-3/PVME-FC : FAILURE in 37:29.44","accounts_in_message":[],"_revision_number":3},{"id":"c28a4a59a000861d1005631f4002f4d6931a744c","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-14 00:20:24.000000000","message":"Patch Set 3:\n\nCould you please review this change?","accounts_in_message":[],"_revision_number":3},{"id":"1565905f390069352137302e00014d09b3239642","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-14 13:30:22.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nThis seems reasonable, and I note the code that requires the cgroup executables was added nearly 7 years ago.","accounts_in_message":[],"_revision_number":3},{"id":"fe10be3e3fb8f7c828a32a1a02535810a7d7589d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-15 00:54:55.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"89be038c141ec7c652a6ad81d4996dd9dfebbad6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-15 00:56:00.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/795009/795009/4/check-not-vote/ext-cinder-zuul/a155424 : FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"9c5a6716cda7f8250e7d3d54212154b571d4dd3e","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-15 01:00:32.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/795009-3/PVME-iSCSI : FAILURE in 39:39.93\n* PVME-FC http://os-logs.tristero.net/dell-ci/795009-3/PVME-FC : FAILURE in 38:28.22","accounts_in_message":[],"_revision_number":3},{"id":"d1416e404ab5f06f920d36642d783c2f1c189239","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:17:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/520873afee6c465989723d15040bd30b : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a54c346b48d4c5c92d071b289334fc0 : SUCCESS in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1463f77a1d43f9b47ee9ae75b6d50a1d469307f2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-15 01:54:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/4/iscsi/ : SUCCESS in 59m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/4/smb/ : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":4},{"id":"136eb002c7125320fada7f1b339b1c7d0e9347f4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-15 01:58:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/09/795009/4/check/iscsi-driver-primera/90a2482/90a2482.html : FAILURE in 1h 01m 42s","accounts_in_message":[],"_revision_number":4},{"id":"62d4c036cea751977689312f37dc4177cf4982f3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-15 01:59:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/4/check/dsvm-tempest-cisco-zonemanager-job/1857187/ : FAILURE in 59m 21s","accounts_in_message":[],"_revision_number":4},{"id":"1c1348abeb96c2cb1327625448fd47b94f50fbc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 02:50:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a52ef116d82e44c3adf69709622c5187 : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddcac57a00034f1eb6630747ed2ac76d : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e09a2972b0e43ceadb0e63bdf01578a : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a77d76835b84de3bb55f5a5521574eb : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e17cfc5ef864b0eb80a3f732bff576b : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7815fb4abd4487eabb5504702c7599b : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167a74dc395e487dbc3b6b39b6de97da : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/10c60e55bd034097b005858ee089306f : SUCCESS in 1h 10m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/703376a448da46dc9fbbc372204d07c2 : SUCCESS in 1h 08m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03b781d8ac4c444b82c78d6c9a4a1f41 : SUCCESS in 7m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4eccb86f1a184f90a3551c179f55c073 : SUCCESS in 12m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bad99f16a6444b13987ad6480a1cc92f : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4c3bb44a8cc491f8de67a123f7636d2 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a1cfa1e8fcc41c898f627babbf697b3 : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2a22c3cd2f2425689105fb4025f1187 : FAILURE in 36m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cee7f1c635bb44d9a569160674d68d1c : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/28df6aa5a0af40a29e6c1da59446bb0e : SUCCESS in 1h 13m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd5e9751711f425983abadf9037dea6d : SUCCESS in 1h 26m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d4bd359487c4c0ab2f2ed61128d5b44 : FAILURE in 31m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/429a420eb19d4735b6c1495647c767b5 : SUCCESS in 1h 12m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4da6714962984caf886939be0e1b43e7 : SUCCESS in 1h 17m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e4ce61c2e68d44799a5dca0bb02fa90a : SUCCESS in 1h 29m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b3440ad2f854899a5cabc5d6e5c2121 : SUCCESS in 1h 53m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dde634e36da4b6b8ad14700341f8452 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":4},{"id":"0949c49c99c36cc336411508c50b9c1a87b800b6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-15 02:52:31.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/09/795009/4/check/DellEMC_XtremIO_FC/526a2ba/DellEMC_XtremIO_FC/292 : FAILURE in 1h 23m 52s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/09/795009/4/check/DellEMC_XtremIO_iSCSI/c9e0e0f/DellEMC_XtremIO_iSCSI/295 : SUCCESS in 1h 56m 59s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"ec8fb9f4c4eb077b973ece4ca70b9715fdd2eb8f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-15 02:56:00.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/530cfff0bf81453393d0ed0dea08645b : POST_FAILURE in 1h 55m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b540ff82c75043d4b4c62c28c68f3034 : POST_FAILURE in 39m 36s","accounts_in_message":[],"_revision_number":4},{"id":"83fc71ed6590b8c7642ec02003c873996d456f3a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-15 02:58:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/09/795009/4/check/DellEMC_Unity_FC/41c3dd4/DellEMC_Unity_FC/301 : SUCCESS in 1h 21m 39s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/09/795009/4/check/DellEMC_Unity_iSCSI/87b87f2/DellEMC_Unity_iSCSI/285 : FAILURE in 2h 03m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"5894c12d7f78f18789854bca79b7b610b0368c29","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-15 03:04:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/2b89bd2e65364ff282c773eaf732a5ee : POST_FAILURE in 1h 29m 46s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/8b7f7dd1c80a49dbaf6a545cf8963dbc : POST_FAILURE in 2h 07m 53s","accounts_in_message":[],"_revision_number":4},{"id":"b442da9d8ae74545861095ff9395cfb2baa785fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-15 03:40:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/09/795009/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8c3b3b8/8c3b3b8.html : FAILURE in 2h 44m 16s","accounts_in_message":[],"_revision_number":4},{"id":"cc62b6c7e4cbe882a45d791481c86a46241433ab","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-06-15 03:59:15.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6d6b31a835a27018da63e8d3e330d3e536a4b304","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-15 04:00:07.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/795009/795009/4/check-not-vote/ext-cinder-zuul/f155819 : FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"7ae35f3f80637baa2a50ba78719a518fe5c8e983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 04:14:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67d460dc72ca4117966ecd86e014351b : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6eccb31101a474986f609eaa152bebd : SUCCESS in 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"12960d99501f703ff36e2f246bc5bffd78cc78b6","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-15 04:23:49.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/795009-3/HPMSA-iSCSI : FAILURE in 1:02:39\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/795009-3/HPMSA-FC : FAILURE in 22:53.06","accounts_in_message":[],"_revision_number":3},{"id":"b932c9d975686af4fc21dfbaaa2222bbc56e6d0d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-15 04:24:00.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/795009-3/Lenovo-iSCSI : FAILURE in 34:14.17\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/795009-3/Lenovo-FC : FAILURE in 32:11.15","accounts_in_message":[],"_revision_number":3},{"id":"8b8682e0135da7f861801643576c27abb606abc9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-15 04:51:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/4/iscsi/ : SUCCESS in 49m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/4/smb/ : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8386a210ff05fd4a0633d4b090c2e9776fdadf58","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-15 05:01:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/4/check/dsvm-tempest-cisco-zonemanager-job/d99f536/ : FAILURE in 56m 22s","accounts_in_message":[],"_revision_number":4},{"id":"b374b6f03d1fc779af1736fe874d71dafed3a65e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-15 05:10:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-4/ns5_nfs : SUCCESS in 4h 14m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-09-795009-4/ns5_iscsi : SUCCESS in 4h 13m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"521343bce934aaf982f62a4781409dcfa84b777a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-15 05:41:34.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b6bb5c50e4ae47b8b9d6bbae99399cc3 : POST_FAILURE in 1h 15m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8f58c110012c479aac63d4cba48a594b : POST_FAILURE in 40m 45s","accounts_in_message":[],"_revision_number":4},{"id":"bd34eba3bf94cd60b031e4b06417f8acc56905ca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-15 06:02:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/c4e533421db840daa27a292368403166 : POST_FAILURE in 1h 35m 40s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/883f0e510e214f7b8aa88cd3178947c2 : POST_FAILURE in 2h 01m 45s","accounts_in_message":[],"_revision_number":4},{"id":"7e7761cfc4b04f5335a52af22550dc41ab828bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 06:43:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3654e65769874a51a294074f84f98d0d : SUCCESS in 11m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc426dea5bd8405a8e397d290b869449 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c32722d252c44642829baeb8613ac21e : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e33f51e2fe44e96b3cb4f401d7a4c91 : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed8ce833f704633940aac82102cb7f2 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d4fe8a46314b33ba614a7b98d53134 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df93b4043944c4199235c68b8d05784 : SUCCESS in 16m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/299e1eae539c46e88ae6a85650ac25e9 : SUCCESS in 59m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3723783f05b0487fa163a7615b3a5359 : SUCCESS in 1h 07m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cebff9b938e54d8fab5567b923d4d67d : SUCCESS in 7m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37d402e109124ac3af495f96239a2e3c : SUCCESS in 13m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/686066d5a59b4031b4facdd401ab459c : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5f9ee951c10445e80a73d0248876c47 : SUCCESS in 8m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b818a2e900ad484b92b22be86307f981 : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/002cc0d5c5e2420196728680ccd5f2bb : FAILURE in 29m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d47b8c9c04974b7f90f8f116b318afd4 : SUCCESS in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2c47fd448f6413091aacdacc75d48b2 : SUCCESS in 1h 08m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/117dc7d19a0f46698ad41f9e44132e2a : POST_FAILURE in 2h 09m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b7b67b37330c4d869cd040ca40270631 : FAILURE in 26m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e46b744792134d778bef0c332c55e0b3 : SUCCESS in 1h 08m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6064eee5675140d0a3d2077f74ef3c84 : SUCCESS in 1h 14m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4cc2c32639548aea6d684b6d73d532e : FAILURE in 2h 37m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a975281bc245439783466b95e3b4ae4f : SUCCESS in 1h 48m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19fd25945aee4fa1afe17ef90e25f72f : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":4},{"id":"55bd9394ef2a017c47940d940e2cb126f2e7e9d2","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-16 00:17:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"14e9ebcb3b6bf25518b9a6d082ddd1e3be8e14e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 00:43:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fc9c59c7fd14998a71d9db4e81e29bc : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4acf4d70caa94ca4990861a70b0bba87 : SUCCESS in 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ebe0bf59864d9e613626de3d9fe9358f589d9cfe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-16 00:50:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/795009/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/da458e2/ : FAILURE in 22m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/795009/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/798a0ec/ : FAILURE in 24m 28s","accounts_in_message":[],"_revision_number":4},{"id":"d2ebc2afa0f465519d199e889e516ec28838f4ff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-16 01:20:56.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/4/check/dsvm-tempest-cisco-zonemanager-job/299e638/ : FAILURE in 57m 17s","accounts_in_message":[],"_revision_number":4},{"id":"f6cbf9f513e7905b70e16ad901898c98f17746fc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-16 01:31:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/4/iscsi/ : SUCCESS in 1h 13m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/4/smb/ : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":4},{"id":"889c75cddc61cedb304f986c6f16bde4a40d9bd5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-16 02:12:49.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1a1b687c779940c8871d1f24734e5745 : POST_FAILURE in 1h 28m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/014c5bedaa08469d89129a126c005090 : POST_FAILURE in 37m 50s","accounts_in_message":[],"_revision_number":4},{"id":"a2d2b0dbc4fe00d7f619f6068ed86a43ce227719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 03:28:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d394702b2a7470eb09a83b6c7b2a3aa : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2498b4babc1f41d0b9911337186b40b9 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283672ea14524482ba6dff7c0dcc7146 : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b637f55f7ff64dd08e85922a8d1ea353 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af9d786cea5543748394bb2369f45f5f : SUCCESS in 12m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e40e665327fc4a54974933fdc5dd77ef : FAILURE in 45m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d297d95f0cc4947b2cddeb33bac5ae4 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/a00ed34cb694475e89c08713406c0d89 : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6f74cfb8ed245c4be304781c9f79b7d : TIMED_OUT in 2h 02m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcbbd329f7894302ab3528780fa9076c : SUCCESS in 6m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a75582866cae454d9a7dbd8deb74940e : SUCCESS in 12m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a95ff1bac782477289dcec2f1e88e687 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3c46b5761ea492ea81009ce9c697b6c : SUCCESS in 7m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9577d67497b4eb4b339c666a41cbb58 : SUCCESS in 7m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b109094a44cf4ebf8d0ed45407db21a0 : FAILURE in 38m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f6089f5b9c5409b9003d86a7e6d5ec3 : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d373d461d1d944359d794845c5fd57ab : SUCCESS in 1h 14m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbaeb8ca8920493b87d65027da3b8c50 : SUCCESS in 3h 09m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08bace61d04341fba09dc91d5d935443 : SUCCESS in 1h 17m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4381ee9ee59847969d16927549567d75 : SUCCESS in 1h 17m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cc5d520070454509b7e6a4c34a19d4dd : SUCCESS in 1h 16m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19e7b6476f2f4d85af182ba615a86c4d : SUCCESS in 1h 10m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e28d471f87544fe5bff46f75e9a8f25c : SUCCESS in 1h 51m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e608de32050b4d13ab33907d894b7d5f : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":4},{"id":"0219161dbdb24b49dd7eef7bc74c4d6f5a3c80a3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-16 13:28:26.000000000","message":"Patch Set 4:\n\nrecheck tempest-integrated-storage (timed out during setup, no tests were run)","accounts_in_message":[],"_revision_number":4},{"id":"d8fa89ff89ad9405f8dc29408e4be7493ef75757","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:57:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5aa9a57a9d5e4d6f9fe2295f9a692a09 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9dd2284d87784ac6bfa2a9256fe27a9a : SUCCESS in 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1bfd0ce0ff5e1c109e9286e2d97caf271712f0d9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-06-16 15:12:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/795009/4/iscsi/ : SUCCESS in 1h 04m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/795009/4/smb/ : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":4},{"id":"0be5283f611443f98ea39b7e0093670db33ef9e9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-16 16:52:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/09/795009/4/check/dsvm-tempest-cisco-zonemanager-job/88b01ae/ : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":4},{"id":"00b91b62173ba001a3e9dff6b77e1f6e4286f19e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-16 16:57:08.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1e4fb9d320d34cb987bef62711991ebe : POST_FAILURE in 1h 57m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/fcbf8fe339044634a43369f1cc6c1d07 : POST_FAILURE in 48m 54s","accounts_in_message":[],"_revision_number":4},{"id":"6b870c7e2812e3c2378e35f3aebb702535d3b1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 16:57:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/22dc8ec81b5f4beeafa631b2ea91e169 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a45e8375ee44117a4bbe137ffd8c480 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee34c1e760c4bd3b7d9c3447de0921a : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7535c29068340309d5462993dea82a1 : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09ff20cd86e1431a997ae1a85cf9a7ef : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee57183ff8da423c833c6a4be6986436 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efd388be37d4db9a26f0ce45915259d : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/dbe696c5e71c49beada51401f59e1d88 : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2b7abdd1f384b6f96e99674db13dd57 : SUCCESS in 1h 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eda00848ca524bbd8deaea356be0cc51 : SUCCESS in 7m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/70b507e62d444b94a1ea9a573c6f5507 : SUCCESS in 12m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b6aaea2e60f49f3a6355658256de780 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2287c2a7ee24358a44565f314ce56e4 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e1c562c5c9f24c49a8cca054dfc9e2aa : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/baee8f03099d425794e0be3a2ef73f3a : FAILURE in 33m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb33d9bf2e044f2cac86d5045b9489e7 : SUCCESS in 7m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9faaf07950ba44f4b407f66f40679492 : SUCCESS in 1h 10m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/772768a68e7b4d3eb4fff0d35021ad1d : SUCCESS in 1h 31m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6bcbbe33d6e148018c30ad05fafc9362 : SUCCESS in 1h 23m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2365206f119e464e9a92fc1185799007 : SUCCESS in 1h 09m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9fd7cfefd3d44bda880c3ff1683ea1d3 : SUCCESS in 1h 18m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/13a891faec0e4b6398fef62019895bdc : SUCCESS in 1h 11m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3aa6629b7a241ecae092fdc524d784d : SUCCESS in 2h 37m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c90c27e9c12d4fdaa29e1458c6092851 : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":4},{"id":"afc59daeb418de94843bffe97aee2cf60af4b194","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-16 18:18:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/09/795009/4/upstream-solidfire-check/cinder-solidfire-iSCSI/fc6888e/ : SUCCESS in 1h 08m 34s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"68d8ea4f25ea5ac067f011d0717f5fed5ee7bbeb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-16 22:46:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/09/795009/4/upstream-check/cinder-cDOT-iSCSI/822979a/ : SUCCESS in 2h 10m 27s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/09/795009/4/upstream-check/cinder-cDOT-NFS/56bb604/ : SUCCESS in 42m 42s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/09/795009/4/upstream-check/cinder-cDOT-FCP/701043c/ : SUCCESS in 1h 49m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"31ac01243fe4d8f9f42ac81b3ec0644fcdc6bb18","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-17 03:53:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/09/795009/4/check/DellEMC_PowerStore_iSCSI/83b716f/DellEMC_PowerStore_iSCSI/158 : SUCCESS in 50m 53s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/09/795009/4/check/DellEMC_PowerStore_FC/a68c246/DellEMC_PowerStore_FC/159 : SUCCESS in 52m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"6f205784b3609aafb69991102781a0ad8a687351","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-18 00:16:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/09/795009/4/check/DellEMC_XtremIO_FC/f723a37/DellEMC_XtremIO_FC/317 : FAILURE in 1h 27m 50s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/09/795009/4/check/DellEMC_XtremIO_iSCSI/15c6d66/DellEMC_XtremIO_iSCSI/320 : SUCCESS in 1h 49m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"b0136549093e0db39e1dea02571cb1ad374f9ac2","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-18 00:54:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4538d5172a31faebb281f9e7a98e2e44a46a7342","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-06-18 15:21:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3c505674ebb4a4e515b157f76ddb8733fb35b878","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-06-20 04:06:44.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/795009-4/HPMSA-iSCSI : FAILURE in 59:59.16\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/795009-4/HPMSA-FC : SUCCESS in 31:40.30","accounts_in_message":[],"_revision_number":4},{"id":"a2196d0e41fb57a51a999479395be149e857b8d1","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-06-20 04:11:25.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/795009-4/Lenovo-iSCSI : SUCCESS in 46:32.73\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/795009-4/Lenovo-FC : SUCCESS in 42:30.49","accounts_in_message":[],"_revision_number":4},{"id":"64396790026810eedf0176019144c726dfd8ab93","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-06-20 12:20:43.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/795009-4/PVME-iSCSI : SUCCESS in 53:11.95\n* PVME-FC http://os-logs.tristero.net/dell-ci/795009-4/PVME-FC : SUCCESS in 53:20.86","accounts_in_message":[],"_revision_number":4},{"id":"e11ad5d4798f353c346cbbd446df5ce5533a2099","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-06-23 00:21:24.000000000","message":"Patch Set 4:\n\nCould you please review this change to move it forward?","accounts_in_message":[],"_revision_number":4},{"id":"314a15531955fe98d5b2b0f2ccd3c58dd609b13f","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-06-24 10:02:55.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"b85186500cc73e2ff0e2a3bfc95c3b7b00c39828","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-07-16 14:41:15.000000000","message":"Patch Set 4: Code-Review+1\n\nAs a follow-up of this issue, a downstream consumer (RDO) was patched to depend on this library (all versions from master to train): \nhttps://review.rdoproject.org/r/q/I60621e3159ad0f2a2829cc494ac5e35b5c529211\nhttps://review.rdoproject.org/r/q/I9fb20e0ba2622045fa38d9901b00b7fb65f0351a","accounts_in_message":[],"_revision_number":4},{"id":"cead9a995ceee018e49bccd30c2111fe5a9c72a8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-16 15:26:32.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\nThis should be backported to stable branches as soon as it merges, because we will be removing this dependency during xena development.  It would be good to have it in the stable branches, though, as a courtesy to operators.","accounts_in_message":[],"_revision_number":4},{"id":"c6031e15945c2d500009041e5f08dc04f3ef09ed","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-16 16:04:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks good to me, and I agree to backport this","accounts_in_message":[],"_revision_number":4},{"id":"849fa7b4b0a36252bb1c38c349666191211558fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:05:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f3eafa2d2938411c75d328e188e1be9ee27d1ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:25:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50b0550f136f41519b44d23765cf7bab : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0f5ab0b9da410086434f8a27fd0463 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef271df5cf304dab9d02c837542526f0 : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c3b748ca8ed44fa83e58c1a6a68bdc4 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af81f32264b343d4be013aa662d307ce : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/2a011f47df1e4fa0acb0d4b2b470d556 : SUCCESS in 1h 04m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be078148ae0d436ba5b59f9b6e0a1b64 : SUCCESS in 1h 04m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3c7790598d41598565aecbe4c6825d : SUCCESS in 8m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22e38895726a407b998352dcacc658e6 : SUCCESS in 1h 10m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74b62b11e39f4523ba0d99c2c49c2f46 : SUCCESS in 1h 16m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a179fc137504e8e94b43e8a34580197 : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":4},{"id":"180c79c4c34fc882767e602b7d48f9ba2d3ed6e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:25:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c3c2fc83383f4dcbe5b2eb007f008e7a1d280a87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:29:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e22fc3ae5cc8408a8272da63a402b4c4 : SUCCESS in 1m 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31df2885f6ef4459b0d79546afc15dfd : SUCCESS in 2m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c27891d9f714d6396f06814e2351a42 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ca1e0bd76beafa20a2240260b1a5116b6ea6bfcd","revisions":{"aba6cd0285636c3d3b8b6632dcd10ea2400642d8":{"kind":"REWORK","_number":1,"created":"2021-06-07 01:15:35.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/09/795009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/795009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/795009/1"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-07 01:11:17.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-07 01:11:17.000000000","tz":540},"subject":"Add libcgroup related packages in bindep.txt","message":"Add libcgroup related packages in bindep.txt\n\nTo use volume_copy_bps_limit option, libcgroup package is reuquired.\nThis change adds cgroup-bin for dpkg based distros and libcgroup-tools\nfor rpm based distros.\n\nCloses-Bug: #1931049\nChange-Id: I179f23d3e6c62d02c069be0dd3d60968a2119568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aba6cd0285636c3d3b8b6632dcd10ea2400642d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aba6cd0285636c3d3b8b6632dcd10ea2400642d8"}]},"branch":"refs/heads/master"},"cf3a75e0fc6a7a83fe62c3b4b7ebe013366383ca":{"kind":"REWORK","_number":2,"created":"2021-06-08 00:45:28.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/09/795009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/795009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/795009/2"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-07 01:11:17.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-08 00:45:14.000000000","tz":540},"subject":"Add libcgroup related packages in bindep.txt","message":"Add libcgroup related packages in bindep.txt\n\nTo use volume_copy_bps_limit option, libcgroup package is reuquired.\nThis change adds cgroup-bin for dpkg based distros and libcgroup-tools\nfor rpm based distros.\n\nCloses-Bug: #1931049\nChange-Id: I179f23d3e6c62d02c069be0dd3d60968a2119568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf3a75e0fc6a7a83fe62c3b4b7ebe013366383ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf3a75e0fc6a7a83fe62c3b4b7ebe013366383ca"}]},"branch":"refs/heads/master"},"9de135b848f1e2693b3ba7bdd52f98a32ea72c84":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-08 00:45:54.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/09/795009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/795009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/795009/3"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-07 01:11:17.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-08 00:45:36.000000000","tz":540},"subject":"Add libcgroup related packages in bindep.txt","message":"Add libcgroup related packages in bindep.txt\n\nTo use volume_copy_bps_limit option, libcgroup package is reuquired.\nThis change adds cgroup-tools for dpkg based distros and libcgroup-tools\nfor rpm based distros.\n\nCloses-Bug: #1931049\nChange-Id: I179f23d3e6c62d02c069be0dd3d60968a2119568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9de135b848f1e2693b3ba7bdd52f98a32ea72c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9de135b848f1e2693b3ba7bdd52f98a32ea72c84"}]},"branch":"refs/heads/master"},"ca1e0bd76beafa20a2240260b1a5116b6ea6bfcd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-06-15 00:54:55.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/09/795009/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/795009/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/795009/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/795009/4"}}},"commit":{"parents":[{"commit":"d5f0e51879c03654db4bf89564695e01af593350","subject":"Merge \"Update IRC info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f0e51879c03654db4bf89564695e01af593350"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-07 01:11:17.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-06-15 00:54:55.000000000","tz":0},"subject":"Add libcgroup related packages in bindep.txt","message":"Add libcgroup related packages in bindep.txt\n\nTo use volume_copy_bps_limit option, libcgroup package is required.\nThis change adds cgroup-tools for dpkg based distros and libcgroup-tools\nfor rpm based distros.\n\nCloses-Bug: #1931049\nChange-Id: I179f23d3e6c62d02c069be0dd3d60968a2119568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca1e0bd76beafa20a2240260b1a5116b6ea6bfcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca1e0bd76beafa20a2240260b1a5116b6ea6bfcd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
