)]}'
{"id":"openstack%2Fcinder~799631","triplet_id":"openstack%2Fcinder~master~I4157bb18f107c940d997358768101c65f78a7f51","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2022-02-20 04:46:05.000000000","reason":"Change was abandoned"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2022-02-20 04:46:05.000000000","reason":"Change was abandoned"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-10 11:53:50.000000000","reason":"Gorka Eguileor replied on the change"}},"hashtags":[],"change_id":"I4157bb18f107c940d997358768101c65f78a7f51","subject":"Dropping explicit unicode literal","status":"ABANDONED","created":"2021-07-06 10:50:18.000000000","updated":"2022-02-20 04:46:05.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b8d7e7d043edf6304f9e892117cb715f85389a97","_number":799631,"virtual_id_number":799631,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-07 05:45:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2022-02-20 04:46:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":-1,"date":"2022-02-10 11:53:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 11:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 11:48:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 11:53:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 12:04:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-07-06 12:08:35.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 12:11:05.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-06 12:47:55.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 12:51:15.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-07-06 12:56:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 13:43:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 14:45:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 15:04:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 16:12:14.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 08:29:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-08 14:14:26.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-07-08 14:17:25.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 15:04:20.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-07-08 15:04:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 15:46:02.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-07-08 16:04:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 17:00:34.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-07-08 17:54:59.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-07-08 17:56:54.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-07-08 18:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-07-12 10:27:38.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-07-12 10:27:38.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-07-12 10:27:38.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-07-12 14:53:34.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-09-27 12:13:50.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2021-09-28 06:24:27.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-09-29 08:56:20.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-12-23 07:43:33.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2021-12-23 11:20:54.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-02-07 03:39:26.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 04:55:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-02-07 05:54:09.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-07 07:03:31.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 08:28:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"bc87c79d7de8d299a145916e6e8faa4b28c77840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-06 10:50:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fd18fd1b8530bf9ed98bb317a0218df7d8fc01f","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-06 10:50:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"74f32f515451aad6c28ffb362de16d895e53fbac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 11:42:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9953f91227946fb8f992e9b9a65257b : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81f9178f3bda408c9a397f15666d9da0 : SUCCESS in 24m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7eb4bf7ac412ea65414f8a659b5b58b0d2ea65ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 11:48:24.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ab035f01c60bdbcd7b2f0a50cf4f30e6d44a004e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-06 11:53:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/799631/1/check/iscsi-driver-primera/42f7796/42f7796.html : FAILURE in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"02b751562738384f63a3268314aa9fe048861eac","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-06 12:04:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/799631/1/check/dsvm-tempest-cisco-zonemanager-job/1cef5ff/ : FAILURE in 1h 07m 09s","accounts_in_message":[],"_revision_number":1},{"id":"70a217e047a884dd58827a4d580f92f0e9db1efb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-06 12:08:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/799631/1/ : FAILURE in 1h 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"29396f249d1170749fe5669a00356c397a9ff0ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-06 12:11:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/799631/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a9c8d78/ : SUCCESS in 1h 17m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/799631/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/df8f1f9/ : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7857f3e6e0f160411732ad08ec877ab09298e53a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-06 12:47:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/799631/1/check/DellEMC_XtremIO_FC/3fffcc5/DellEMC_XtremIO_FC/440 : SUCCESS in 1h 20m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/799631/1/check/DellEMC_XtremIO_iSCSI/30b0f67/DellEMC_XtremIO_iSCSI/434 : SUCCESS in 1h 48m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7d4ddac11b85d5e011f6914206eedd92cbac7ea","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-06 12:51:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/1/iscsi/ : SUCCESS in 1h 02m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/1/smb/ : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7d2919389e1d0e6ab38843134efa4936a35a6175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:56:42.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2de727b646204038ac68d8c7992b7e41 : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef6ff2d9fdbd4edfb8662106d302163b : FAILURE in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee360b5daa4f43b6bd2519c366377002 : FAILURE in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/838403ce2b64430aaebe398dee862077 : SUCCESS in 12m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23dedfbce2ef4b22ae4808a42500c40a : SUCCESS in 13m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d825225b74d401da0f1b424e1259459 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efa2aea2a5846f7a067420c175f2a83 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fabd8322cc7a4d29940f918f399bbbf6 : SUCCESS in 1h 06m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bebfb41a24fa4944b3dd8c331fb69735 : SUCCESS in 1h 25m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2200c5e1179144bbb987d7fa6441be74 : SUCCESS in 6m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/467ee19ce95044ef9103833f84c8609b : SUCCESS in 14m 07s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d82162e5e6ab46358c0761070d81b96b : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6837fcb7dd7d460a8817cda13a4756fc : SUCCESS in 8m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5002457b0a3a485e8d8b2a25a988a362 : SUCCESS in 7m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/633afd0e9bd14e398d0b3ed99405314c : FAILURE in 44m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3933599f51474e96a27756fd75ddf34b : FAILURE in 8m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4153d00b1dd142f3b270563b98501759 : SUCCESS in 1h 16m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5004a25176584a46a2705ddbbdf0936e : SUCCESS in 1h 49m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c1e70e0c23a24533823c1e2da13a7268 : SUCCESS in 1h 49m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/53264119c240418d8c8b3287e421fe5e : SUCCESS in 1h 12m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/03d5a9e0208248c28d252fe3f3a01560 : SUCCESS in 1h 31m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4ccf0cf030041b5b0927171782e92fe : SUCCESS in 1h 24m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92ea9aed20c54ed3b1174f7b213dbfe7 : SUCCESS in 2h 02m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1450bdf4fb1d4fd1b0036c5c4d146663 : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fb0e6957283ac61bff99daf200c4dc88566c4d4c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-06 13:14:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5c41d1c/5c41d1c.html : SUCCESS in 2h 23m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b0fd21c0098b05cd12b22b522ec3e213cca10f2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 13:42:04.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cf81602f3b43a6cc03465a185ef6124984d0c7e2","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-06 13:43:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/1/20210706195023/NEC-FC : SUCCESS in 2h 30m 03s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/1/20210706195023/NEC-ISCSI : SUCCESS in 2h 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fcf3cd4c8707d1a6efcb9a779ffd40757df14059","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-06 14:45:52.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/799631/799631/1/check-not-vote/ext-cinder-zuul/8ce7ca7 : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d7b4c54cb79e2eac2fdb4bb66834490ebf688eeb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-06 15:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-1/ns5_nfs : SUCCESS in 4h 10m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-1/ns5_iscsi : SUCCESS in 4h 13m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"cc54e9c42d387de445f65e0d7fd3d461b55b4d11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 16:02:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"be462ef340cac0dc5336ade933c1e527b32d7cad","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-06 16:12:14.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/799631/1/check/tempest-dsvm-full-focal-py3/acdf26d/ : SUCCESS in 1h 52m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/799631/1/check/devstack-plugin-ceph-dsvm-focal-py3/4b87c7c/ : FAILURE in 49m 05s","accounts_in_message":[],"_revision_number":1},{"id":"515b72d46c0202f141efdfb8536e7c4f1db4fa08","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 17:08:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"341c339d466cda332ca5d6100fc0ef6a9a38c1e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-07 07:39:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"380fcf73949b536633a13ba913413abbfdf1296a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 08:01:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a25f0986d884fd6987a8ee219ac8209 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e499fa95c814b4c9675a93fcf6d5438 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4affb83e0132c00e99a91293eb5d8d0b8270d30","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-07 08:15:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c50bd9e/c50bd9e.html : FAILURE in 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"af5875e2ded4ed5585758a1dbeb6aee3c0a4e109","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-07 08:36:32.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"127e9b55d40e0118c5be5d51ce6c35ef7da0fab2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-07 08:46:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/799631/2/check/iscsi-driver-primera/f044f4e/f044f4e.html : FAILURE in 1h 05m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5eb7b6c6fe2784ae451422652e30568fe5840e36","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-07 08:47:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/799631/2/check/dsvm-tempest-cisco-zonemanager-job/cf79796/ : FAILURE in 20m 37s","accounts_in_message":[],"_revision_number":2},{"id":"82e741c07feb3451c8640c3ca3eba27972bf5a94","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-07 08:57:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/799631/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/177f3d2/ : SUCCESS in 1h 16m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/799631/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3dde8b2/ : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2a68680dd241df53436217a0b95fa8727f2ce33c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-07 08:59:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/2/iscsi/ : FAILURE in 1h 19m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/2/smb/ : SUCCESS in 49m 15s","accounts_in_message":[],"_revision_number":2},{"id":"bfd50280fd4379b878aa74c66349a201b2d05059","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-07 09:10:54.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/624800827fc14f45ba05fa5d024a2af3 : POST_FAILURE in 1h 26m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"002691b25ce34cfc6abafc487e3561758843d59d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-07 09:14:03.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/2/SPDK : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"e1024e1b855c145daf06a56d6d590fdbc357e995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 09:41:12.000000000","message":"Patch Set 2: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7ef9c153e720461da47b027b5d04fc26 : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6af98c6f062042cb89d87a64cd40d60f : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f42aeefc8547cdb5b391f95f3a9381 : FAILURE in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fe8009f7f23417292fa825f3f3ba0fe : SUCCESS in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8fb045c78bf4234a59ab0218e3fe010 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35a0fdd74f11434d8ed797ce90cbe1cb : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b17d3b97d194a799d1a2a01f8282723 : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/26bd44cf642c42f89a514f58f8a2bb69 : SUCCESS in 58m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e7449a0aef849f58d1b1fe6a4ea8c6c : SUCCESS in 1h 05m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75c7f57cf6b44465828bde398de26bf6 : SUCCESS in 7m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3f488ff25bef49d39fa3ffcdefc95bdc : SUCCESS in 11m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc40afd2e5e94c1589cb548cfc6bf102 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7eab05c9c6fa487f97a0c4cbdcaf5cf3 : SUCCESS in 7m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a34650496bb40c9890575416a075dda : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/670fa2c919d3403d95c58233e13648e3 : FAILURE in 30m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2448644458ac4b6bbe111cbd8a513157 : FAILURE in 8m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0cae856a4084ed1be909e619193ca73 : SUCCESS in 1h 24m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d87ce25456a943c5ae52986b2cb4c6c2 : SUCCESS in 1h 24m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/493d3f82a17e4a1c922e9215c92ac1d7 : SUCCESS in 1h 18m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5cae626bc3248da82f37c082c763b1a : SUCCESS in 59m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad93759735e94d6f999de256a1fc9889 : SUCCESS in 1h 02m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/601878b069154d10a59a7838c48e0233 : SUCCESS in 1h 10m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cf45a4ad68e49faa43f996595c64908 : SUCCESS in 2h 00m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1c6d9a9cfce4c008b8284397a77d495 : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":2},{"id":"eb35ba3d9b6bb43ddde76f30c0a39ab5a25cb87c","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-07 09:44:00.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ad6450df8168d828c3ab8b4c1a947ae685d51400","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-07 09:44:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/799631/2/check/DellEMC_XtremIO_FC/ba3a39c/DellEMC_XtremIO_FC/448 : SUCCESS in 1h 23m 34s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/799631/2/check/DellEMC_XtremIO_iSCSI/ee39a7c/DellEMC_XtremIO_iSCSI/442 : SUCCESS in 2h 04m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"b6af0d730349a949410dff117f52b6b35e3886d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-07 09:45:44.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/2/SPDK : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"f86de189189777b731af683c54ecf32965eda140","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-07 10:15:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/2/20210707163934/NEC-ISCSI : SUCCESS in 2h 34m 31s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/2/20210707163934/NEC-FC : SUCCESS in 2h 36m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d7578d0fc6959a09585f6609a9f73481b05b6308","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-07 10:30:03.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/799631/799631/2/check-not-vote/ext-cinder-zuul/46bb13e : FAILURE in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"af42e28f50ffe3fd57553020cdbec90cc504adf7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-07 10:36:23.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a9ad0e6578c1ea0d533e7b7070c58c7a3db7ffbd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-07 11:54:22.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-2/ns5_nfs : SUCCESS in 4h 10m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-2/ns5_iscsi : SUCCESS in 4h 14m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"dfd2975d1d7670fd88bd6593511347394567e53a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-07 12:51:59.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"efb39dc44612282b2bb58e1717daab87e3e8904b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-08 05:40:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"53026b2122ad3340a3f330c856d30ef1361195cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 05:59:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd3cdca30595425e9c320832c1050e31 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad24ef08f18e48939c103edd4f224112 : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c60127bf5372a3e0d3ca59dd641bb00e02e6eb4a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-08 06:30:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/3/iscsi/ : SUCCESS in 49m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/3/smb/ : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":3},{"id":"973a4217b7218b0b7e808b04e6a7f731545f2c93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-08 06:38:56.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"648a19c346bb850a4a208d1f4155b21b6bbd7659","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-08 06:51:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/799631/3/check/dsvm-tempest-cisco-zonemanager-job/4403b52/ : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":3},{"id":"852dabe75d2809ecc3fea34faf42dae0cd92c319","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-08 06:57:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/799631/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c9a2aa9/ : SUCCESS in 1h 12m 15s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/799631/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/70e3ab7/ : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c7e05721680b1d337b3a1037ad4d810e971e733d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-08 07:04:58.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f2d43d900eb2472d9dd764706ed15df3 : POST_FAILURE in 1h 20m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"1b8b0910e7e67e6491902172359620b6225fe438","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-08 07:10:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/799631/3/ : FAILURE in 1h 25m 12s","accounts_in_message":[],"_revision_number":3},{"id":"d1cd99d1fbbfb8470b5d5529fd771ef7fdbc30d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 07:42:34.000000000","message":"Patch Set 3: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5d27d405dc6e40f182fe8515f26eab6a : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd59d140fd6e432f9718170b2e474ddf : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e788650f41425099b67fef0e64da50 : FAILURE in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a812d9bc3f104868ade40245d0a59f64 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32b20f68fdbd496486722ff0e6577dbd : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6af32a12deaf4c2da44b428045b2609c : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b042fbbc52ee47c1a98530db115b1927 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/9df1d01e1ade4f0e8125f4897ccd9848 : SUCCESS in 1h 05m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9036adc2e4b24451ba70128a1d5e7f51 : SUCCESS in 1h 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b8904a2b19c4f20846b57e6f9e6ff1e : SUCCESS in 6m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/091c46c675e8446da76cca31a74d56d8 : SUCCESS in 12m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06d517f054da4ed3ab91b12643b4b445 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92dcda68b3904d5588404e0d7c4f81bd : SUCCESS in 6m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1b21f817ca94f2ba5d167ebfedc3c37 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1d9cce4c381f4010a078c490da9d576f : FAILURE in 37m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cad2c3466064e128fa495236869ee94 : FAILURE in 7m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a788c5e259b4d77a37616411c59b52c : SUCCESS in 1h 27m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/27f0ebd31bea4eed866b7b25da94205e : SUCCESS in 1h 27m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/96a1907ccb834b46ae3cd2d0df42a473 : SUCCESS in 1h 44m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4e0d631a3560438c8355b0898de4ee65 : SUCCESS in 1h 16m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b5584b79be248d682eca3b32e5f719f : SUCCESS in 1h 03m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/02e6d9405cc8427687a6dc47399482f8 : SUCCESS in 1h 08m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07c4e7c737764ede82c137b764b4cbc4 : SUCCESS in 2h 00m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4317cd127e043f48126174fae7f2667 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3079252f9d43c5f179abc69a7f53b853bf7d408d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 07:57:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0df3bd4/0df3bd4.html : SUCCESS in 2h 16m 04s","accounts_in_message":[],"_revision_number":3},{"id":"b56e34495aa7c3c5ebdc1e72086eeba72e5de661","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-08 08:01:53.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"da400ef699877e6935581df4f07f3a64a1c1f244","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-08 08:29:22.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/799631/3/check/DellEMC_Unity_FC/479c275/DellEMC_Unity_FC/393 : SUCCESS in 1h 22m 44s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/799631/3/check/DellEMC_Unity_iSCSI/22dcfa1/DellEMC_Unity_iSCSI/369 : SUCCESS in 1h 50m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c441dc52dbbc2a6d1a99d8573ca9b20aa0bd116d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-08 09:06:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"859d57b7551ed7588fc0e22570c102fde1cfea3d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-08 09:08:00.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"936496153200a69c2acce9f7b9bf7f93b26f3b6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 09:28:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e940af98678489fb4cb2719f5809f1a : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/700e46ee22014eba923e8a6c4636e1c3 : SUCCESS in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"84f641591cc910907c0552206b932af164704474","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-08 09:57:36.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-3/ns5_nfs : SUCCESS in 4h 16m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-3/ns5_iscsi : FAILURE in 53m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"3ed65f768f988432644caf6e6adc32613ea26bf0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-08 09:58:40.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/4/SPDK : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"c9936ce1a88f89b880bec45f5a909e41848b5efb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-08 10:04:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/4/iscsi/ : SUCCESS in 56m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/4/smb/ : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d5ba3f97cb8e91428aff60055772f0bc8f274cf7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-08 10:12:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/799631/4/check/dsvm-tempest-cisco-zonemanager-job/cb67867/ : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":4},{"id":"5b02df8483d748f5489352f22f6d9429c6b5cdab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 10:13:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/799631/4/check/iscsi-driver-primera/2017632/2017632.html : FAILURE in 1h 04m 37s","accounts_in_message":[],"_revision_number":4},{"id":"c4ee98303ee9574a776542d506b79c795079c935","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-08 10:26:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/799631/4/ : FAILURE in 1h 11m 58s","accounts_in_message":[],"_revision_number":4},{"id":"3b9520e4e222dd3169f81f0077bac8f8e333f21e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-08 10:55:58.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c8df80f6aa5f66db5d0d07b743fad7bc675fa325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 11:07:54.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d2df0533a684f43870865980505afad : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acfa1e4f346b4819b8082f6fa35afe85 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f8b5c98ce044f198e9b03f6b88b3b8 : FAILURE in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6110799f1474ef1b38bfba83ce602b5 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99409c4257ef4d3c94f5636d8e5fd50f : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/210ced91acbc40d29475c2cd30aa69b8 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6aab146a5b24a82ab3e80afdc94e03f : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/47c5fec26bff476a883f7371d615c4fc : SUCCESS in 1h 12m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8f63217109d43038631228fc926b593 : SUCCESS in 1h 02m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d7d41706df40889edda8477e32be34 : SUCCESS in 6m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2fa272055aaa4bc7a5a1fa475ac8bee1 : SUCCESS in 12m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec0904fc131d443e984c1cfe8a3676ba : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5fdf1095f684e4095a4ef86ae0b6d02 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53c9f2992d2644d18ad9ab4ff078743f : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c4f15b2d001647a8832c53b2be390ecc : FAILURE in 29m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d4cab0b8c79d4a589b92e4d4c97bfb5b : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ffd71e0f7160455dbb7bc8516f86e4ca : SUCCESS in 1h 05m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ff9a521482649d791da2c501e1e3098 : SUCCESS in 1h 22m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1f9cfbaad6444b2a186cdb16790187e : SUCCESS in 1h 24m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b444571f33745e2987b4d03139f8cfe : SUCCESS in 1h 06m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71d8b7a1c05c4838b889ca9e8ef27c36 : SUCCESS in 1h 30m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c703528fcb684b7bafc1b0ccafb612b3 : SUCCESS in 1h 08m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c3a7e5c42fb4426982514d9ed4ad855 : SUCCESS in 1h 59m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee038a499c3a4e6f85b503142e09f055 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":4},{"id":"1afa77fc4e635c38abb9920a4fc9bc5fe981c4cd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 11:21:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e76c9cc/e76c9cc.html : SUCCESS in 2h 12m 57s","accounts_in_message":[],"_revision_number":4},{"id":"97fe36d41a6d400b48c307a99018056e02f39d84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-07-08 11:23:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"22e77876644b446d9e4c67350fae91262cca8a1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 11:44:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/015ac5accf5f4cb982d4687811aa672d : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c76906f8295a48ad960806afc635dadd : SUCCESS in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c215c258924454814f7b53a9cf4556f11332368a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-08 11:49:03.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"b7ba4a9dd28135d139ced08d657cae85e808f8f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-08 11:49:36.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/5/SPDK : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"d1b86be30079cdbc9287e422d9c014db46d15315","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 12:26:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/799631/5/check/iscsi-driver-primera/db2ef8a/db2ef8a.html : FAILURE in 1h 01m 44s","accounts_in_message":[],"_revision_number":5},{"id":"d8c66ada3905191776ff611ea3d88fc5f6f05193","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-08 12:41:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/799631/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5080488/ : SUCCESS in 1h 16m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/799631/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/10398a4/ : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":5},{"id":"f1b692b4e9eb761ac17ed0feeb991be42ec2d986","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-08 12:43:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/5/iscsi/ : SUCCESS in 1h 19m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/5/smb/ : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":5},{"id":"c4e36fcca6032f94d4dfd7d5aa5c5d94637b3a24","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-08 13:18:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/799631/5/ : FAILURE in 1h 25m 46s","accounts_in_message":[],"_revision_number":5},{"id":"b28eb4f8178c9571af6ca69e84c93550d9d1b151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 13:20:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/760d7645da954835a443560863b3b1f1 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/680bdf2a1abe43e1b5eea787fd35301c : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121be0329ba14a1794a57e8e4e8d824d : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d51fe90253345a4ab7c9f1d47603302 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c527d2f0d1149959889c183a9f89a82 : SUCCESS in 9m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02c20540802a464bb83fa77a8b1fd950 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c58650f072463c87411a4a656f0264 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/625b673d09dc403691676542d3625ae7 : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aca10e56023640ec8504a9b21652c846 : SUCCESS in 1h 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dd615b045d94ac9943a416dd387e41f : SUCCESS in 6m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1d5655a4c18e4e0495fec5aba6f79bde : SUCCESS in 12m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/62c7323fbb7d4d43bbac6d8a15392895 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c31edd7af3df4f688fb7e2a164e90275 : SUCCESS in 7m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/720ea943d5634a55bf0ac85cd2a1f1fe : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c4d8cf689d58469faaee329ec65ba87f : FAILURE in 36m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b9ed3f6668743888c5025a14d0c0f5c : FAILURE in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f6cac241bb049268e5386e4a5f91db3 : SUCCESS in 1h 10m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6470d9e39d1f4fa2b8f2eec8a5d9d81b : SUCCESS in 1h 26m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b2c1f449ea449d48dd9d7cdf8945cac : SUCCESS in 1h 23m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6aa71932a2e24b6fbf39cb0c2e31b19d : SUCCESS in 1h 07m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9bd0946e6cc548b386be7802c6797fa1 : SUCCESS in 1h 25m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/24f86817218346d596db6ff6e94b5dfe : SUCCESS in 1h 11m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/664e6d7a667044a08ac0d94179a719ba : SUCCESS in 1h 55m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d67ef627ad7a418980274cbd8b04b20c : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":5},{"id":"9899286e4df39392efe8184010d4f95cda6394c2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-08 13:41:49.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-4/ns5_nfs : SUCCESS in 4h 28m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-4/ns5_iscsi : SUCCESS in 4h 34m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"637b603188fdb1d6e7bbd5f0d2059cb98f274c0f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-08 14:14:26.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/799631/5/check/DellEMC_PowerStore_iSCSI/0330077/DellEMC_PowerStore_iSCSI/218 : SUCCESS in 53m 48s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/799631/5/check/DellEMC_PowerStore_FC/115c503/DellEMC_PowerStore_FC/219 : SUCCESS in 50m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"48f7fb46a96a7c3d295c30b493c8d576fb88d5ee","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-08 14:17:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/31/799631/5/upstream-check/cinder-cDOT-iSCSI/847029f/ : SUCCESS in 36m 20s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/31/799631/5/upstream-check/cinder-cDOT-NFS/9e96dfd/ : SUCCESS in 35m 17s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/31/799631/5/upstream-check/cinder-cDOT-FCP/417033b/ : SUCCESS in 56m 09s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"48cf24defb1f5f6572f19590bcacf4a66637b8b5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-08 14:45:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c15caf7/c15caf7.html : SUCCESS in 2h 24m 11s","accounts_in_message":[],"_revision_number":5},{"id":"04cf06187050b852a082eb63b3dee7e62014352b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-08 15:04:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/31/799631/5/upstream-solidfire-check/cinder-solidfire-iSCSI/ad7a5b5/ : FAILURE in 1h 42m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"661238ac75e211d187091820672ef7b1d3440fa0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 15:04:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/31/799631/5/check/DellEMC_VNX_iSCSI/8ebf848/DellEMC_VNX_iSCSI/248 : FAILURE in 1h 14m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/31/799631/5/check/DellEMC_VNX_FC/b632c4b/DellEMC_VNX_FC/266 : FAILURE in 53m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"24f64d08ca8ee713c1c051ac993319cf2db305fb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 15:45:00.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/799631/5/check/DellEMC_XtremIO_FC/8ca0a42/DellEMC_XtremIO_FC/466 : FAILURE in 1h 18m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/799631/5/check/DellEMC_XtremIO_iSCSI/0a21d8f/DellEMC_XtremIO_iSCSI/457 : SUCCESS in 1h 49m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"3205cb7d4481090d4678d6d9fea5b9fd7ea5fae2","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-08 15:46:02.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/799631-5/PVME-iSCSI : FAILURE in 9:29.24\n* PVME-FC http://os-logs.tristero.net/dell-ci/799631-5/PVME-FC : SUCCESS in 47:20.16","accounts_in_message":[],"_revision_number":5},{"id":"2abef4a71d680fe3d859ea28842c609c072804bd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-08 16:04:38.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/31/799631/5/check/DellEMC_PowerFlex/544e66f/DellEMC_PowerFlex/287 : FAILURE in 1h 33m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"c185e937705c06018b0c6265a91c0ea28c8289db","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-08 16:43:56.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-5/ns5_nfs : SUCCESS in 4h 35m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-5/ns5_iscsi : SUCCESS in 4h 43m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"fce83c2ef6b8dddc4a7263eecbab5fdc6f3dafaf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-08 16:55:36.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/799631/5/check/DellEMC_Unity_FC/360b558/DellEMC_Unity_FC/398 : SUCCESS in 1h 20m 31s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/799631/5/check/DellEMC_Unity_iSCSI/d3c7efa/DellEMC_Unity_iSCSI/374 : SUCCESS in 1h 42m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e960bb5971974d5fce2fff15ddd2aefba00241ae","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-08 17:00:34.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/799631-5/Seagate-iSCSI : SUCCESS in 40:25.78\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/799631-5/Seagate-FC : SUCCESS in 37:28.63","accounts_in_message":[],"_revision_number":5},{"id":"5e0e1e3b0fefc0be6037fb3e0b335f4253887d0c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-08 17:54:59.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/799631-5/Lenovo-iSCSI : SUCCESS in 49:09.87\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/799631-5/Lenovo-FC : SUCCESS in 45:26.61","accounts_in_message":[],"_revision_number":5},{"id":"376f180dfbe521e5e249e1726a77272439444c87","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-08 17:56:54.000000000","message":"Patch Set 5:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/799631-5/HPMSA-iSCSI : FAILURE in 1:01:06\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/799631-5/HPMSA-FC : SUCCESS in 33:27.35","accounts_in_message":[],"_revision_number":5},{"id":"ce23c097a430a5f9bea5f57a8f9bd08078ff0a56","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-08 18:00:05.000000000","message":"Patch Set 5:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/799631/5/fc : SUCCESS in 0h 46m 44s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/799631/5/iscsi : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":5},{"id":"2ff30300b907999ea6f729e9fa5037dcfd616375","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-08 22:27:04.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/799631/799631/5/check-not-vote/ext-cinder-zuul/cc92638 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":5},{"id":"fdf006b18b33d8bd3abceb2d8eae6a077b4f8a0b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-09 04:36:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/5/20210709110000/NEC-ISCSI : SUCCESS in 2h 33m 35s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/5/20210709110000/NEC-FC : SUCCESS in 2h 36m 54s","accounts_in_message":[],"_revision_number":5},{"id":"d3120eed230f6a5770ada9f841710139a19ed75e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-10 12:13:16.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/5/SPDK : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"5f08c35425728e3bdff7d1648e3b0c139283c5aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-11 05:09:56.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/5/SPDK : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"0f4ecf17a1b5c7cd7ccfaf151cdd8395ca83137f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-11 21:25:50.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-5/ns5_nfs : SUCCESS in 4h 10m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-5/ns5_iscsi : SUCCESS in 4h 11m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"81a8400fb661e7540f815d3772f567a08e43b77f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-11 22:27:15.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"7f197fdcaca7bc9264aff6fdcfbce554bbf940f7","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-07-12 14:53:34.000000000","message":"Patch Set 5: Code-Review+1\n\nUnit tests are clean, I don\u0027t see any reason why u\"..\" cannot be removed from the code base.","accounts_in_message":[],"_revision_number":5},{"id":"5e291b04ec423cdb70f456c2bf1eca2007e1ee42","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-23 04:21:44.000000000","message":"Patch Set 5:\n\nping","accounts_in_message":[],"_revision_number":5},{"id":"8d2a02ac3800701f401d5c007c05df6378d1b7ea","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-26 02:51:09.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-5/ns5_nfs : SUCCESS in 4h 29m 33s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-5/ns5_iscsi : SUCCESS in 4h 30m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"9f3e572f7ecae15b549749536854c6d98d93368f","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-27 09:19:22.000000000","message":"Patch Set 5:\n\nHi core reviewers, Could you please help me in reviewing and merging the patch.  Thanks","accounts_in_message":[],"_revision_number":5},{"id":"753eab47ff495d88b1fd44d7c2ff84095023facd","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-09-27 12:13:50.000000000","message":"Patch Set 5:\n\nHi,\nIts been 2.5 months since the last patchset (PS 5) was submitted.\nMultiple files are updated.\n\nSmall suggestion: It would be nice if \"rebase\" is performed.\nSo that, later on there are fewer merge conflicts (if any).","accounts_in_message":[],"_revision_number":5},{"id":"1bd30cb34c1dec798c445fdeaa5a6dec7712ef6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-28 03:25:28.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"35353de9452f28716c69225a0d4256f2ffe85f7f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-28 03:48:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/799631/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e89e14d/ : FAILURE in 21m 51s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/799631/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4372f2e/ : FAILURE in 21m 05s","accounts_in_message":[],"_revision_number":6},{"id":"259cac063a4a4eebae9849bf4681f8c74f0b9e4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 03:49:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c88365973da34d4e882c5aa625a642bd : SUCCESS in 17m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52e9896c8e5e410fb26abf4f28c6c111 : SUCCESS in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d468007bb6b5552d4082a425824a9f307e9d94b4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-28 04:05:03.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_Unity_FC/5269863/DellEMC_Unity_FC/970 : FAILURE in 39m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_Unity_iSCSI/1da0aad/DellEMC_Unity_iSCSI/837 : FAILURE in 30m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"309cffd2296a1edb6fbcedfcf488cb540bb0c2ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-28 04:15:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9d2ebf1/9d2ebf1.html : FAILURE in 48m 01s","accounts_in_message":[],"_revision_number":6},{"id":"aed519ce8e703a617335cb6f15058f033180a28a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-28 04:21:36.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-6/ns5_nfs : FAILURE in 51m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-6/ns5_iscsi : FAILURE in 55m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"431f683dd03f1df9283ac032f4f65e9d34672b0d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-28 04:29:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/799631/6/check/dsvm-tempest-cisco-zonemanager-job/b6dc422/ : FAILURE in 57m 25s","accounts_in_message":[],"_revision_number":6},{"id":"de0b02b27da38392b471f3ed8c73742e346a2925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 04:34:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6fb512641d8545a78e6378d9dc7a7f28 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd32828eb72425aa9e10fa345ffc358 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21aaf2ebc13946d8849ff7a9b49eb6b1 : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a365d19afddd44cba0a91ac9caa62fef : SUCCESS in 10m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7f6acaa1e0e4026aa96b1f32be36f59 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fedca50310504c8bb4c70db6b10486ee : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/cacd194dc7c54a1aa004afbbd192d7a4 : FAILURE in 24m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8240e4fbe759488585ed1a3a0e169f01 : FAILURE in 23m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb00e225a494c4ba2f38518a82370f4 : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/36ea55d3c6a4476a9a1a93d2710c5474 : SUCCESS in 12m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd986f9923e944d7bc7550412a8a73a8 : SUCCESS in 6m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01882b087a4548f68bbc15e11c895bd6 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/753033dc73cf4e8fa2cef2b82ee3f1c1 : SUCCESS in 6m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb517bfe56b54b608cc90ea28f19aff4 : SUCCESS in 6m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca4625541aeb436f9a606b09d7852806 : FAILURE in 26m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d67be4e1c69f425f8ce9749183cfb2a7 : FAILURE in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/658acc57b32840c8a658eb8cd82b2a85 : FAILURE in 27m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6b9f658e1d0d465baf728e6b5198a220 : FAILURE in 34m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4bd75f9582dc4141a1ed557bf2980a93 : FAILURE in 26m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fe662e4d3e0b4d21aaed29bcd1a238bb : FAILURE in 34m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b11b64fce632451c96d3c7a439e43668 : FAILURE in 23m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a32e23170acb4bed95825525f039f7d4 : FAILURE in 25m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f247730119f0431b9167082248adb40c : FAILURE in 28m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19e396e831eb48ab81a62ab70a01979f : FAILURE in 55m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f32afbdcb18840df94287d3f59c099e8 : FAILURE in 24m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95880b18c7234262bcb10e3b4ccedb6a : POST_FAILURE in 26m 02s","accounts_in_message":[],"_revision_number":6},{"id":"f157a1da314ebc27f01250b1d3d844fa2bf14e34","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-28 04:34:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/31/799631/6/check/iscsi-driver-nimble/bc8864b/bc8864b.html : FAILURE in 1h 08m 08s","accounts_in_message":[],"_revision_number":6},{"id":"776016ead5c9645e716a534e276416859485f93d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-28 04:36:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/799631/6/check/iscsi-driver-alletra-6k/3580da4/3580da4.html : FAILURE in 1h 09m 29s","accounts_in_message":[],"_revision_number":6},{"id":"89859d2447c571e392e38a6dfd63e16463453e79","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-28 05:52:25.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b93e01a5a7c8c4b9bc68e681db34134aad98ac0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 06:19:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bba4d5eedd164a169f40430cf92b474c : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4327e406ea6d4383946e2ec079f7c733 : SUCCESS in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c2da9384969dec27cd8fcd72c0187c3ae79c422b","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-09-28 06:24:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/42ab4a6/index.html : FAILURE in 31m 19s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/7f16c15/index.html : FAILURE in 31m 22s","accounts_in_message":[],"_revision_number":6},{"id":"ae8a711413f51c6dfb0a66e7e735479373cceeff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 06:33:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2f3b2633b2db4396a11b4bf26967b0bf : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46e81d22214446aaa9844793d7af87b1 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298945a58c794dfcb5c4e96fd70a614d : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a84b08f509d7411981ea635985e68673 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b17c36f2fefd4c5ebd2f30efa24a38f4 : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd82af50645f4b16b9d35b05c7142ebf : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a3723fea5d9845bc8af53528bb822f92 : FAILURE in 22m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de7d896d47634b2b8373396d896f7210 : FAILURE in 23m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f6699bc82724800b7231eb596149785 : SUCCESS in 7m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8b5323bf3918457ba7e44da3a6f80592 : SUCCESS in 12m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2173c10d82d34c6cb780eb4562ee7309 : SUCCESS in 6m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6ac61e5515840de9152aabec5bb4d2a : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c353df74612491da1f9edda0739eb93 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a3624342915c4b7c85aef06e22acf81d : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5748afbc8484a758071f09036bd4537 : FAILURE in 34m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b278b5eb71d4922ae833fdad761c416 : FAILURE in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec4bd9ba1a7a4f39b8bd0c53133b5ba5 : FAILURE in 26m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/979980acba314a798f10704a76722185 : FAILURE in 30m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dbfee13dcedd410c96752b396522fcf8 : FAILURE in 25m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/45739addc6684b6b8022c6bb6b45c0c2 : FAILURE in 35m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0bde3e87168e45ad9fe751332ee4e941 : FAILURE in 24m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0e46c477895a491c832975926f9d2e9e : FAILURE in 23m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f41cab74b16449f09505bac8c0088452 : FAILURE in 23m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8afc05d2068e4dcfb2479ee3880dd014 : FAILURE in 27m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b10f1d86b3448a1826384532bda66eb : FAILURE in 22m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19b7390a153f4f019d7dc8722027cba4 : POST_FAILURE in 23m 36s","accounts_in_message":[],"_revision_number":6},{"id":"d3f1dc514073e6a3d0d99954d1dd1090a029f57c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-09-28 06:45:37.000000000","message":"Patch Set 6:\n\nIn below link, please refer note \"latest failure: all devstack based jobs\":\nhttps://etherpad.opendev.org/p/cinder-xena-ci-tracking","accounts_in_message":[],"_revision_number":6},{"id":"ec9349109de03ca4288758438445520731be1f8a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-28 07:25:10.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/6/iscsi/ : FAILURE in 3h 59m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/6/smb/ : FAILURE in 3h 56m 44s","accounts_in_message":[],"_revision_number":6},{"id":"24c86b5c527b18dde0f84b36c8954d4ce1270da5","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-28 07:57:57.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e In below link, please refer note \"latest failure: all devstack based jobs\":\n\u003e https://etherpad.opendev.org/p/cinder-xena-ci-tracking\n\nOk thanks","accounts_in_message":[],"_revision_number":6},{"id":"684b10473a1186fff7f83e7fc14faff7de2208b2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-28 08:02:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/799631/6/check/dsvm-tempest-cisco-zonemanager-job/6bbb9e3/ : FAILURE in 57m 40s","accounts_in_message":[],"_revision_number":6},{"id":"fe61915097edef3568419cf4b785d839437c300f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-28 10:31:01.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/6/SPDK : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3062e29bfb0f17e2838fa3b4b942ec7a95dee5da","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-28 17:06:11.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/799631/799631/6/check-not-vote/ext-cinder-zuul/1b18dfc : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":6},{"id":"abe75a9f49f3182d7e299deb3f15ba3145c244e6","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-29 03:26:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"dbcea3ce0e1029427f25c5fcc8b3ca4a96d70b39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 03:50:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e79a0caf47ad42058ad3234be9ec3978 : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7e1e4fcc6064f599ff6e195e8566adb : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"28f069ce7a86f4f67a473877fd433c116a9ad24e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-29 03:54:52.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/6/SPDK : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"a48e7d00335bc85990061b58136d2877acb4b56c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 05:41:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3dc3676518c84c808f5370b95d9c30ff : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3570c7d99c7446e1b15960a3155627e3 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/008fe523a1d74fad821fdaddf9e7b85f : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5705e31d29fe4001bd319d1bbd1a1da5 : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7307cd6adb6e484d82504066619b4a64 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653deb8d3e814c50ac51c3b3d1915d5d : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e143254d96c042dcb0131c13e1b92cea : SUCCESS in 55m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2768c3784151404bbbaee96f51b8acbf : SUCCESS in 1h 03m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8450cb87a1d34040844b6f510c5df9e8 : SUCCESS in 7m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fca9c0727b30480398151b5f54b6ffa3 : SUCCESS in 13m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1dde6070ec634eeba8dead1ec17cb93a : SUCCESS in 6m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9853e559156344099ba11fbef45931ae : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05998b307ce943bebafaffb36f05d80d : SUCCESS in 6m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b02c4a326fb4405890f534bbfe2ac9b4 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1e7602256594c95b3075504918676b3 : SUCCESS in 36m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b80cdd2db1b547f3b6d3ed537def9a01 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f794874134664b6fbaa0a78a468bbf97 : SUCCESS in 1h 11m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/46b41e08380f41d4aea1567f774b7f00 : SUCCESS in 1h 42m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/472f889d19d2433f8cef280398130643 : SUCCESS in 1h 27m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5b1c7a5e91548239da4d39975eb660a : SUCCESS in 1h 07m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2431b82b49cc42c5b4bda7e4cf55159e : FAILURE in 2h 13m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/df181f75326f453dbeb5c9b2a8588db7 : SUCCESS in 45m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df35f691ac79451597d958e1478869a4 : SUCCESS in 1h 09m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5417a2cdf6a49f5b1e3438a8dffeb97 : SUCCESS in 1h 54m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/440e641a883244ce812faa173d0d3a42 : SUCCESS in 40m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab7f2068bccf42c090a18446a260ceb9 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":6},{"id":"ef83de27e1f1b178c39b0a4e4235c80f206e5044","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-09-29 05:50:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/34aa009/index.html : SUCCESS in 1h 18m 29s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/45bde7f/index.html : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":6},{"id":"70ddf726a443a68d8ba31183f0323e1b941b391d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-29 05:59:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/31/799631/6/upstream-check/cinder-cDOT-iSCSI/050cdd6/ : POST_FAILURE in 16m 01s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/31/799631/6/upstream-check/cinder-cDOT-NFS/95ac1f2/ : POST_FAILURE in 16m 01s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/31/799631/6/upstream-check/cinder-cDOT-FCP/6330bb2/ : POST_FAILURE in 16m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"aca3f84aa8228ff094dfd2b7290741b85a819959","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-09-29 05:59:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/31/799631/6/upstream-solidfire-check/cinder-solidfire-iSCSI/4d2324f/ : POST_FAILURE in 16m 00s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"5b3e03e7b266d9c34e7ab30699f7989c7691e11b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-29 07:16:57.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_Unity_FC/d85c0d2/DellEMC_Unity_FC/976 : FAILURE in 4s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_Unity_iSCSI/2edfb50/DellEMC_Unity_iSCSI/843 : FAILURE in 1h 28m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"461a954f81439ab5ae3552b836e80ee711490ae7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-29 07:24:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_PowerFlex/dca92b8/DellEMC_PowerFlex/701 : FAILURE in 1h 40m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"b748c27ecb15dbbe89ae47c6401cd840b7fd882b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-29 07:24:22.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/6/SPDK : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"5d90755e03d89c8cc5636fabc0e9f92b894254c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-29 07:55:23.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/6/SPDK : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"71ff5c63eca6f3f8a86055dfb1f45b38287e3c77","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-09-29 07:55:53.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/799631-6/PVME-iSCSI : SUCCESS in 42:40.12\n- PVME-FC http://os-logs.tristero.net/dell-ci/799631-6/PVME-FC : SUCCESS in 42:42.02","accounts_in_message":[],"_revision_number":6},{"id":"99925263f381ec3957fe0cb09da4c5d7419a0777","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-09-29 08:13:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/799631-6/HPMSA-iSCSI : SUCCESS in 55:51.38\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/799631-6/HPMSA-FC : SUCCESS in 35:07.08","accounts_in_message":[],"_revision_number":6},{"id":"62201828313c478bc8739f9d094df20addfdc6bb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-09-29 08:56:20.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_SC_FC/b97cb5f/DellEMC_SC_FC/618 : FAILURE in 47m 22s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_SC_iSCSI/a41a1a5/DellEMC_SC_iSCSI/586 : FAILURE in 1h 42m 27s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"f699a52bda7c289dc5dcb7cf69e7dfd48fa49bf2","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-09-29 08:59:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/799631-6/Lenovo-iSCSI : SUCCESS in 1:06:16\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/799631-6/Lenovo-FC : SUCCESS in 43:49.06","accounts_in_message":[],"_revision_number":6},{"id":"6794edaf5741c86ba95a55efcc0416040ad662f5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-29 10:10:20.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_VNX_iSCSI/afb67fc/DellEMC_VNX_iSCSI/625 : FAILURE in 2h 45m 38s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_VNX_FC/9e958d9/DellEMC_VNX_FC/614 : FAILURE in 2h 01m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"fc3cc0a29e40ee432afbf9ae9061c96aa8c8dc76","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-09-29 12:38:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/799631-6/Seagate-iSCSI : SUCCESS in 3:04:28\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/799631-6/Seagate-FC : SUCCESS in 35:28.55","accounts_in_message":[],"_revision_number":6},{"id":"fd379158b350268d7a21abc3c59887e0fe82d93d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-30 06:11:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"d7d3b5a843a919535b54ed42f74cb8747e2c25fa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-09-30 11:09:01.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_PowerStore_iSCSI/82bc22b/DellEMC_PowerStore_iSCSI/None : RETRY_LIMIT\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/799631/6/check/DellEMC_PowerStore_FC/0e8c248/DellEMC_PowerStore_FC/725 : FAILURE in 1h 08m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"020d366763189a72c4fa89944f1e7e715cc2bd6a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2021-09-30 14:37:24.000000000","message":"Patch Set 6:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":6},{"id":"5ae606210b02cef320eac3956688c125c35de253","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-30 15:56:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/799631/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2ab521b/2ab521b.html : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":6},{"id":"f9b14f49a97346e26f927af3f9f3b336d85ee8b0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-10-01 01:27:48.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/799631/799631/6/check-not-vote/ext-cinder-zuul/afaa940 : FAILURE in 1m 52s","accounts_in_message":[],"_revision_number":6},{"id":"9c3159349e128480e0d59c22e395d4a7022a37c8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-03 23:12:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/6/20211004051353/NEC-ISCSI : SUCCESS in 2h 55m 41s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/6/20211004051353/NEC-FC : SUCCESS in 2h 58m 55s","accounts_in_message":[],"_revision_number":6},{"id":"ad435d13eb2e517ee2175aee2272093b39baccb7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-12 12:38:50.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/6/SPDK : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"4d867a22e8acf7ad197073fd431a04fe5272906c","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-11-23 07:14:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"24419f69402f72998089594a8556c95958594c1a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-26 00:20:34.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-6/ns5_nfs : SUCCESS in 4h 31m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-6/ns5_iscsi : SUCCESS in 4h 33m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"a1cb63e3a5426346402748c8594e6b259544a8c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-26 01:01:21.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/31/799631/6/SPDK : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"762bff586da62c219467dbca996663c931abd84e","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-01-05 04:38:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f778b58cf42685814ede80432888180c2bbc82cb","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-02-02 16:07:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"12b596fd9c0d6d18c07da793f40172ccdbba2342","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-02-07 03:34:17.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a388f7509fdd08ff94d16e9ca47a42b74d557196","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-07 03:39:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/799631/7/check/dsvm-tempest-zadara-driver/9ec875c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fbf7417fa4b02c91e0685e99b78d8415a85b6227","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 03:50:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_XtremIO_FC/6e01bf8/DellEMC_XtremIO_FC/656 : FAILURE in 15m 11s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_XtremIO_iSCSI/454b835/DellEMC_XtremIO_iSCSI/666 : FAILURE in 15m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"855238f64670ee7caa60d8a92939317426617b29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 04:02:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e804bece06b04355b435b7bbb3d0a3ea : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de2cbcf1546246859cf059a01cb413ff : FAILURE in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"974b5169ed32dfd859a7d6c477ae2372d3a5b7a0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-07 04:54:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/31/799631/7/check/dsvm-tempest-cisco-zonemanager-job/b3e4883/ : FAILURE in 1h 12m 07s","accounts_in_message":[],"_revision_number":7},{"id":"f5b006004821fc92a5b3a9ef0c12040412282a6c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-07 04:55:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/799631/7/check/iscsi-driver-alletra-6k/c73c3a9/c73c3a9.html : FAILURE in 1h 19m 39s","accounts_in_message":[],"_revision_number":7},{"id":"df56c0147b023849eb1ddd542b0b3388047baef1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-07 04:57:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/799631/7/check/iscsi-driver-primera/9e3e56a/9e3e56a.html : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":7},{"id":"71f56a09599c1fd4423a5ebbf6f202d66c04d05a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-07 05:01:26.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/799631/7/iscsi/ : FAILURE in 1h 26m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/799631/7/smb/ : FAILURE in 53m 02s","accounts_in_message":[],"_revision_number":7},{"id":"f122ab8566a986fc8c5fb3d439faa6f415cf8bad","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-07 05:29:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/799631/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/99fbb27/ : SUCCESS in 1h 53m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/799631/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ce5f2ea/ : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":7},{"id":"920da9b9bddb0457fe974088eb7ce85d85ddd5d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 05:45:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/218b71b44a304cd89924f2e90162f0e3 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca32b1c8b1441a188a410361ff6c1fd : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91a5a75da60c4c1582042652f9ad2a88 : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f6d6fe36754232ae1f18c23ec24885 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8cf04a53c54fac86ef8208089a2dca : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/5287a01d0ad44507a235521025ca20d0 : SUCCESS in 50m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/146cc4bc283940cc9dcc8d23fae8722e : SUCCESS in 1h 02m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0c4c031597b47d69bbd5b38e0f80d4c : SUCCESS in 44m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e3f955951904d30a762bdf85fc2710f : SUCCESS in 7m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f3a99d597d64955957e5a24b4d82984 : SUCCESS in 13m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1fa2107053cc470faa4d123e92cc233d : SUCCESS in 6m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f4ebbdeddbd4de188bbe199319f01e1 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50a90140e1e240dab7a09fc04ecfba22 : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d003f4e867a54ddaa1c45da8cb7a8f3e : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ef16010123f498ca20c9163033cdd21 : SUCCESS in 37m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca42a38df3bf45629f861906992a4d47 : FAILURE in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fbf9df11a82464ba16084442fc549c3 : SUCCESS in 1h 12m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/796b438f305f4a7a90285bd7e86bf3c3 : SUCCESS in 1h 18m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/33d1db4d4d304401b1d7ff4310042b92 : SUCCESS in 1h 28m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ae9588b6f214aaea644337733889be8 : SUCCESS in 1h 00m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/01a0680d5dfc4b44adfb9bbc0fef8cd0 : SUCCESS in 1h 12m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/506c62f63d6b4ba589fe12922320f03b : SUCCESS in 35m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ee7f83309884dea81d75249223358e4 : SUCCESS in 1h 10m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/075fa098ab714a499a46583615f22732 : SUCCESS in 2h 06m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f68eb26759874b12a9cf765614f5c060 : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":7},{"id":"f30fcfafd448a9564b5988306509c53b65d678af","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-07 05:47:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/31/799631/7/check/iscsi-driver-nimble/eda1d0e/eda1d0e.html : FAILURE in 1h 16m 47s","accounts_in_message":[],"_revision_number":7},{"id":"bc0865692ddbe0648a2aa8217f54bce6ce37b381","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 05:53:26.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/799631/7/check/tempest-dsvm-full-focal-py3/77286e1/ : FAILURE in 1h 28m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/799631/7/check/devstack-plugin-ceph-dsvm-focal-py3/37c57c5/ : SUCCESS in 2h 14m 05s","accounts_in_message":[],"_revision_number":7},{"id":"a76c99d798b36e6ee4e27c7d4b09fc0590cf62bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-07 05:54:09.000000000","message":"Patch Set 7:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/07_02_2022_05_45_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"24cc5fcb07385cd9da476b882bb6a762bcf85046","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 05:58:27.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_PowerFlex/b3609ea/DellEMC_PowerFlex/371 : FAILURE in 12m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"f04d5cc26a0e1bcdbfca0672daca4880cf514bff","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-07 05:59:51.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_PowerStore_iSCSI/9a9d1b1/DellEMC_PowerStore_iSCSI/321 : FAILURE in 13m 19s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_PowerStore_FC/4fbea3e/DellEMC_PowerStore_FC/338 : FAILURE in 13m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"176253380739fb205b4a6074582c7933ebc8bdd3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 06:01:46.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_XtremIO_FC/94aa258/DellEMC_XtremIO_FC/659 : FAILURE in 15m 14s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_XtremIO_iSCSI/51ad478/DellEMC_XtremIO_iSCSI/669 : FAILURE in 15m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"ae0cb8b167f76312da22b8cb00294303fb894e4b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 07:03:31.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_PowerMAX_FC/18600ec/DellEMC_PowerMAX_FC/283 : FAILURE in 53m 13s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_PowerMAX_iSCSI/a4cc07e/DellEMC_PowerMAX_iSCSI/296 : FAILURE in 27m 22s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"a97adf89415b67b0a25db03964cd782212c4e12a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-07 07:32:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/799631/7/ : FAILURE in 1h 28m 00s","accounts_in_message":[],"_revision_number":7},{"id":"e5638c55ae4107de6ba26a16107f73e204c10f5c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-07 08:03:06.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-7/ns5_nfs : SUCCESS in 4h 28m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-799631-7/ns5_iscsi : SUCCESS in 4h 18m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"d18dc2b950922be799d6ffcb860c7af69fb6a0d5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-07 08:28:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/7/20220207151046/NEC-V-FC : SUCCESS in 2h 00m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/7/20220207151046/NEC-V-ISCSI : SUCCESS in 2h 17m 51s","accounts_in_message":[],"_revision_number":7},{"id":"57ade41779fbef476f98abbb9a14aae7a04221be","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-07 08:33:49.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/799631/7/upstream-solidfire-check/cinder-solidfire-iSCSI/1cada89/ : FAILURE in 35m 01s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"77a4d216b13da3bd32754a887045d04935fa715a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-07 09:41:13.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_SC_FC/113b023/DellEMC_SC_FC/305 : FAILURE in 1h 07m 51s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/31/799631/7/check/DellEMC_SC_iSCSI/b2cbb9c/DellEMC_SC_iSCSI/301 : SUCCESS in 1h 44m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"8ec7e5ecc49fb2e63d700c8f8b11b72d6b375423","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-07 10:13:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/799631/7/upstream-check/cinder-cDOT-iSCSI/a4d023a/ : SUCCESS in 2h 07m 37s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/799631/7/upstream-check/cinder-cDOT-NFS/342ea67/ : SUCCESS in 36m 14s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/799631/7/upstream-check/cinder-cDOT-FCP/830629a/ : SUCCESS in 57m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"812f31e4aee43c471548904568f7c1e7c9973908","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 02:20:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/7/20220209075930/NEC-ISCSI : SUCCESS in 3h 19m 28s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/799631/7/20220209075930/NEC-FC : SUCCESS in 3h 20m 34s","accounts_in_message":[],"_revision_number":7},{"id":"1c232575a33c9f2919fbdb991a638d7d25b1a0e6","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-10 11:53:50.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6dc527b24969797110741b73bdfeba0ff80c1c0d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-02-20 04:46:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5fa2ffd0e9ac7a994a307b02ecb713af353521cc","revisions":{"90a49c08b5bf9efb9b7e964a2d875bc2f67ff0c2":{"kind":"REWORK","_number":1,"created":"2021-07-06 10:50:18.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/1"}}},"commit":{"parents":[{"commit":"a7e98dba5b0bc84a2642952b66962d2e645c64f3","subject":"Merge \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e98dba5b0bc84a2642952b66962d2e645c64f3"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:50:12.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/90a49c08b5bf9efb9b7e964a2d875bc2f67ff0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/90a49c08b5bf9efb9b7e964a2d875bc2f67ff0c2"}]},"branch":"refs/heads/master"},"2e27398d1d70c2f7197beb2de5c70d07d27e7e08":{"kind":"REWORK","_number":2,"created":"2021-07-07 07:39:25.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/2"}}},"commit":{"parents":[{"commit":"a7e98dba5b0bc84a2642952b66962d2e645c64f3","subject":"Merge \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e98dba5b0bc84a2642952b66962d2e645c64f3"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-07 07:34:07.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e27398d1d70c2f7197beb2de5c70d07d27e7e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e27398d1d70c2f7197beb2de5c70d07d27e7e08"}]},"branch":"refs/heads/master"},"1806775522bbf13425abfa2e28242eabfa73bcd5":{"kind":"REWORK","_number":3,"created":"2021-07-08 05:40:05.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/3"}}},"commit":{"parents":[{"commit":"a7e98dba5b0bc84a2642952b66962d2e645c64f3","subject":"Merge \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e98dba5b0bc84a2642952b66962d2e645c64f3"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-08 05:39:12.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1806775522bbf13425abfa2e28242eabfa73bcd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1806775522bbf13425abfa2e28242eabfa73bcd5"}]},"branch":"refs/heads/master"},"26d6652f0abe2e7194b61241c7945a3fea1ea74e":{"kind":"REWORK","_number":4,"created":"2021-07-08 09:06:47.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/4"}}},"commit":{"parents":[{"commit":"a7e98dba5b0bc84a2642952b66962d2e645c64f3","subject":"Merge \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e98dba5b0bc84a2642952b66962d2e645c64f3"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-08 09:02:04.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26d6652f0abe2e7194b61241c7945a3fea1ea74e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26d6652f0abe2e7194b61241c7945a3fea1ea74e"}]},"branch":"refs/heads/master"},"f9959ffc6b7eeefd9d0107a2e14bc60be1999e6b":{"kind":"REWORK","_number":5,"created":"2021-07-08 11:23:01.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/5"}}},"commit":{"parents":[{"commit":"a7e98dba5b0bc84a2642952b66962d2e645c64f3","subject":"Merge \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e98dba5b0bc84a2642952b66962d2e645c64f3"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-08 11:22:12.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f9959ffc6b7eeefd9d0107a2e14bc60be1999e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f9959ffc6b7eeefd9d0107a2e14bc60be1999e6b"}]},"branch":"refs/heads/master"},"c7d5ea20a1dedd0d748d9fdeb782a14a4f28151f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-09-28 03:25:28.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/6"}}},"commit":{"parents":[{"commit":"51fd9aeed4d35c480edb301dcd378930cea31ae4","subject":"Merge \"Delete attachment on remove_export failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51fd9aeed4d35c480edb301dcd378930cea31ae4"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2021-09-28 03:25:28.000000000","tz":0},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7d5ea20a1dedd0d748d9fdeb782a14a4f28151f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7d5ea20a1dedd0d748d9fdeb782a14a4f28151f"}]},"branch":"refs/heads/master","description":"Rebase"},"5fa2ffd0e9ac7a994a307b02ecb713af353521cc":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-02-07 03:34:17.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/799631/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/799631/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/799631/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/799631/7"}}},"commit":{"parents":[{"commit":"d6788fb24d993f84955588ebe4da4b7e10467672","subject":"Merge \"Rework backup process to make it async\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6788fb24d993f84955588ebe4da4b7e10467672"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-07-06 10:45:25.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-02-07 03:34:17.000000000","tz":0},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I4157bb18f107c940d997358768101c65f78a7f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5fa2ffd0e9ac7a994a307b02ecb713af353521cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5fa2ffd0e9ac7a994a307b02ecb713af353521cc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
