)]}'
{"id":"openstack%2Fcinder~800268","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~Ia4b1c303a057fe70cf88bdbbebc1d4f62474f011","project":"openstack/cinder","branch":"stable/wallaby","hashtags":[],"change_id":"Ia4b1c303a057fe70cf88bdbbebc1d4f62474f011","subject":"Abort volume creation when encryption spec is invalid","status":"MERGED","created":"2021-07-09 14:37:16.000000000","updated":"2021-08-02 17:24:42.000000000","submitted":"2021-08-02 17:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800268","meta_rev_id":"359f6865f239cf19b8b7e041d1d163c53aecafcd","_number":800268,"virtual_id_number":800268,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 17:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-08-02 16:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-21 13:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-08-02 16:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 15:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-09 15:54:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-09 16:51:59.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-09 17:15:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-09 17:22:44.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-07-09 18:10:34.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-09 19:25:51.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-07-09 22:59:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-10 07:47:50.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-21 13:35:25.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-08-02 16:04:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"ddb5add1af71374090341968e4ed50b6180aef15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-09 14:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e505c0cc3691a850ce3b33ef460d2d611837148b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 15:04:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0aa3359685045ab8da8988e4ff20591 : SUCCESS in 15m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd3da2007aec47949a062f9f4b203219 : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"beb77575268de797a31f750db0ba0580b6e42707","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 15:54:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/68/800268/1/check/iscsi-driver-primera/f8aa9d5/f8aa9d5.html : FAILURE in 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7c8e4265a20e5be2831108abffdf6aaf11a97a98","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-09 15:59:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/68/800268/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/32cc98e/32cc98e.html : FAILURE in 52m 14s","accounts_in_message":[],"_revision_number":1},{"id":"59681b64ccdff3c62329c67cc9766c3701d92b81","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-09 16:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/800268/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0afb32b/ : SUCCESS in 1h 52m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/800268/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b60681d/ : SUCCESS in 1h 50m 23s","accounts_in_message":[],"_revision_number":1},{"id":"131524b74f67ae6e31a67b242c4c51f788e94efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 17:15:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb40742a104643a08e47e669220a9d56 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/893af95839c04b73b6cd663e7ef96742 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/442b0f2dd1ab48cc9c963509f7083087 : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2eeff5ce29c4e5dbdc865d9905f044b : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce1c1b780ce4cf09d5dc81310a81c21 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/945fc3c8593c43dbb3306fcb4538caef : SUCCESS in 1h 05m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/660918b3a3c845fda7ba07fa9ea719d6 : SUCCESS in 1h 03m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb5e19db55fa4a7a8c8022f0d85c001f : SUCCESS in 13m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/54debcc297954f6481c05d334b32366d : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e7dafa867ca4bbf8444ee773de8fd31 : SUCCESS in 6m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2f2e049fd6634807aa09ad63beec8f61 : SUCCESS in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bcf92f15668d4d84ab7ebb8133e36e3b : SUCCESS in 43m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ced8e9f61b0455da38b3fdca3d0bbad : FAILURE in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c37025ccd9d54747b18eeb4ae27e7288 : SUCCESS in 1h 11m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08ca5f1e40c14404b8bdddf51bcb3956 : SUCCESS in 1h 31m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92dc0cbbf4c146febf4c59322069a2c3 : SUCCESS in 1h 44m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00f6f5cff0524cee8f8ba471ed608763 : SUCCESS in 1h 07m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8e9c84fa9d1446299fe5a5723355814 : SUCCESS in 1h 24m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8880a4c4a5f4eb1b5fa75f3a0e1cdad : SUCCESS in 1h 25m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43946615bab0402bbdb36f0dafce4902 : SUCCESS in 2h 32m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f527ba82cb84aed9b8a8c64e1ec3454 : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ed8985467eb80dbfae7a6e8cfb547cf61a03f75a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-09 17:22:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/68/800268/1/upstream-solidfire-check/cinder-solidfire-iSCSI/36a779a/ : FAILURE in 5m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"753b7f98d5d3d92cdb781512c96da58575676bd7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-09 18:10:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/68/800268/1/upstream-check/cinder-cDOT-iSCSI/4e2f57a/ : FAILURE in 38m 29s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/68/800268/1/upstream-check/cinder-cDOT-NFS/412e7bc/ : FAILURE in 40m 05s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/68/800268/1/upstream-check/cinder-cDOT-FCP/4b77b1d/ : FAILURE in 53m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"45d0e96d4387d80722c17c3250178aac0fbdefd1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-09 19:25:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800268/1/iscsi/ : SUCCESS in 1h 00m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/800268/1/smb/ : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c969e6788da027f3059d8f76c1906f81c73aaa2d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-09 22:59:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-800268-1/ns5_nfs : SUCCESS in 5h 9m 33s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-800268-1/ns5_iscsi : SUCCESS in 5h 12m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e0daf8162b374cc8b7479711000130f94b590c8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-10 07:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/800268/1/20210710140356/NEC-FC : SUCCESS in 2h 37m 35s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/800268/1/20210710140356/NEC-ISCSI : SUCCESS in 2h 43m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1685adc2e56035311e5397723b861db5628861fd","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-07-21 13:35:25.000000000","message":"Patch Set 1: Code-Review+2\n\nMerged in master, code LGTM. looks appropriate to backport.","accounts_in_message":[],"_revision_number":1},{"id":"2606f4685e05b20677a361181fc7f75f86563741","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-02 16:04:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI agree with Rajat.  Although this is patch 1 of 2 for the bug, it allows cinder to fail earlier when the encryption-spec is invalid, and is independent of the second part (which would prevent invalid encryption-specs from being created in the first place).","accounts_in_message":[],"_revision_number":1},{"id":"313971674ff131a7e2edee649cd797afb5a289bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 16:05:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ab204722e13d754fe159e9a3070fe95abdd16ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 17:22:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8158367011e44ad3af21c77fd64fed38 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5408e4f526714b11bd6c2e928765c1c8 : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b13f281b67fc4aa8849c6fd4967bc67e : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68968e08b634423e9bd8eb5aaa8b3c04 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/fa82f0165149491a93d37d2a7e188d3a : SUCCESS in 1h 06m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c7d050bc9e24690ba3f4567a3cc917e : SUCCESS in 1h 09m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/443222d350b3428ea6979dc660a5817c : SUCCESS in 1h 08m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72f81265ae33442aae84fe455a76a584 : SUCCESS in 1h 16m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b7d78a47c08409b9d93877947a65d6b : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"27c6acbd26129f7cc3ea89429049b1362c0b6d7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 17:22:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"359f6865f239cf19b8b7e041d1d163c53aecafcd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 17:24:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bb529ff68184c2d8b1ed349a1cd4531 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0297964207251ef0094cf61d97dbcbabd9a706d3","revisions":{"0297964207251ef0094cf61d97dbcbabd9a706d3":{"kind":"REWORK","_number":1,"created":"2021-07-09 14:37:16.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/68/800268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/800268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/800268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/800268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/800268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/800268/1"}}},"commit":{"parents":[{"commit":"fbc1ccbeacc993436044d04331a21c12d5ac19b8","subject":"PowerMax Docs - Wallaby documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbc1ccbeacc993436044d04331a21c12d5ac19b8"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-05-04 14:06:55.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-09 14:36:31.000000000","tz":-240},"subject":"Abort volume creation when encryption spec is invalid","message":"Abort volume creation when encryption spec is invalid\n\nWhen the encryption spec is missing the cipher field,\njust fail the request rather than sending an invalid\nrequest to Castellan and letting the Barbican server\nreject it.\n\nThis happens when an admin had previously created an\ninvalid encryption spec.\n\nSimilarly, check for a malformed cipher field in\nimage_utils decode_cipher, such as \"aes\" instead of\n\"aes-xts-plain64\".  An invalid cipher field like this\nwould previously cause a failure during volume creation\nin some volume drivers.\n\nPartial-Bug: #1926630 (1/2)\nChange-Id: Ia4b1c303a057fe70cf88bdbbebc1d4f62474f011\n(cherry picked from commit 0a1cc1a2d42d17ee5fcf3d273d543b598b033318)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0297964207251ef0094cf61d97dbcbabd9a706d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0297964207251ef0094cf61d97dbcbabd9a706d3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
