)]}'
{"id":"openstack%2Fcinder~800703","triplet_id":"openstack%2Fcinder~master~Iea1f29ce269107cd6422c40ca2e42ddb7354486b","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-02-07 19:52:17.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2022-02-07 19:52:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea1f29ce269107cd6422c40ca2e42ddb7354486b","subject":"Remove various unused code","status":"MERGED","created":"2021-07-13 18:02:01.000000000","updated":"2022-02-07 21:05:25.000000000","submitted":"2022-02-07 19:52:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800703","meta_rev_id":"af72b91e90b7730073bd4fe80437210838bccdd4","_number":800703,"virtual_id_number":800703,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-07 19:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-04 16:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-02-07 21:05:25.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-07 21:05:25.000000000","post_submit":true,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":-1,"date":"2022-02-07 21:05:25.000000000","post_submit":true,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-04 16:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-13 18:25:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 19:18:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 20:59:05.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-07-13 22:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-13 23:38:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 00:25:39.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-07-14 01:28:30.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 01:44:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-14 05:07:04.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-07-14 08:57:35.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-07-14 09:12:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 10:39:27.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-07-14 11:08:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-14 12:15:43.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 16:16:30.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-07-14 17:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-07-17 04:24:06.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 20:25:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-27 20:40:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 22:01:08.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-28 12:27:53.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-12-15 17:23:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-15 20:21:56.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-12-15 22:31:07.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-16 06:00:07.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-16 11:21:51.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-17 14:21:41.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-12-22 15:47:53.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-01-11 15:56:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-01-17 11:37:51.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2022-01-24 15:26:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 16:47:30.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-01-24 17:45:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-24 18:15:49.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 19:21:46.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-01-24 21:53:08.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-04 16:24:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-04 23:06:42.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-02-07 14:44:25.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 14:57:40.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"CC"}],"messages":[{"id":"823b245eb03764fb4a36e358e5dc9758bd434cd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-13 18:02:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c707b74cb89088efab3322198aca54f8e1a912c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 18:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c418282eac864b24add5c2a4c2e62dbe : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87c2129ec63f4689934e4643da7dac76 : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00b0df97acf8786b7260ccc99f25f531e0c2702a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-13 19:18:41.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/800703/1/check/devstack-plugin-ceph-dsvm-focal-py3/209d9d0/ : FAILURE in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f60bdb99af799de49bfd3e408bb8c19433015336","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-13 20:59:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800703/1/iscsi/ : SUCCESS in 49m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/800703/1/smb/ : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a122b9bec7b642148463f89b4f6d54526c781c16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 22:39:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9fde7410ff4fe4bdf9bffcfcb6a27b : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be054369d4c140b19623559a08793f8e : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a599bef53924ce8aff66d443e0fd409 : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb076ea1169e43e594d9c5df32bebc06 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f9f1082e27402a8e860afdd6828478 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/439425c3c04745a1b350119b837fbb87 : SUCCESS in 1h 07m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0de7f6d11ed3405aa5d7e63ccea517e4 : SUCCESS in 1h 08m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cc57cdb4712042f0821f2437ef295370 : SUCCESS in 14m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3eadcaf75e2b40448fb14ebbb35c9373 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da5480f61afb44c5bf5882ae1aa0c597 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8dde0e3059a8489294635a418f62cf2d : SUCCESS in 7m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0e448ce4b5ef4548bef4a98ae4dd7e08 : FAILURE in 26m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fea57d5d01c545c59ce81f401bbb7ecb : FAILURE in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77ee754c14704706b17d7ed3ae10e706 : SUCCESS in 1h 08m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e782d123b23d4d5095459143be10e826 : SUCCESS in 1h 16m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9708e93e33643f78822e69b768a9d14 : SUCCESS in 1h 29m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/916e97921d164dba8db963078b3aa69c : SUCCESS in 1h 02m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef50c5e8919c49c3b440505f245e3d97 : SUCCESS in 1h 05m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b2cb9fda19454323ba0679838f7e0fb7 : SUCCESS in 1h 10m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aae1bf8a77a84caf8b41241f3c5c2a8b : SUCCESS in 2h 35m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2f10e16f81047af9813b27594971963 : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":1},{"id":"486c8336b52dce4be4fb95414369d849a0ee73fe","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-13 23:38:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/800703/1/upstream-check/cinder-cDOT-iSCSI/b795a1b/ : SUCCESS in 35m 09s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/800703/1/upstream-check/cinder-cDOT-NFS/4d1dc18/ : SUCCESS in 36m 47s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/800703/1/upstream-check/cinder-cDOT-FCP/dc6ec45/ : SUCCESS in 57m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ca1c25b4d0a78ade78a4aed4c4437ceec1b7f5a7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-14 00:25:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/800703/1/upstream-solidfire-check/cinder-solidfire-iSCSI/48a79d7/ : FAILURE in 1h 44m 56s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"290e09f4a8fefb5667bf87f7e46362109f37396b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-14 01:28:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/800703/1/check/iscsi-driver-primera/6f09257/6f09257.html : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8afed89a2a4460391e50f1d86d50c1c01ba59cfc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-14 01:44:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/800703/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/48a78e7/ : SUCCESS in 1h 36m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/800703/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/14aae2b/ : SUCCESS in 1h 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9c071f5ea1741a933b5e29e698c4d3b1d0f7700f","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-07-14 05:07:04.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/800703-1/PVME-iSCSI : FAILURE in 9:26.75\n* PVME-FC http://os-logs.tristero.net/dell-ci/800703-1/PVME-FC : SUCCESS in 46:46.17","accounts_in_message":[],"_revision_number":1},{"id":"819ab9250cc50948bd68fea9c6c8d3374bb550ef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-14 06:36:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/800703/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1bebf6b/1bebf6b.html : SUCCESS in 2h 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"01c79230bd8fe92e250cb3f69a6d4ba39c166041","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-07-14 08:57:35.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/800703-1/Seagate-iSCSI : SUCCESS in 41:19.41\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/800703-1/Seagate-FC : SUCCESS in 38:00.21","accounts_in_message":[],"_revision_number":1},{"id":"a6ddd6e8af370fd4f7e342f052d64319fd5979ad","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-14 09:12:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-1/ns5_nfs : SUCCESS in 5h 20m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-1/ns5_iscsi : SUCCESS in 5h 24m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b6cf722ed068a2d44244efe2696083af8c0fd53","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-07-14 10:39:27.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/800703-1/HPMSA-iSCSI : FAILURE in 1:00:02\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/800703-1/HPMSA-FC : SUCCESS in 30:55.58","accounts_in_message":[],"_revision_number":1},{"id":"321042c8ac32b324d79aaab3b08a970737ff00fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-14 11:08:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/800703/1/check/DellEMC_Unity_FC/79387e2/DellEMC_Unity_FC/444 : FAILURE in 8m 35s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/800703/1/check/DellEMC_Unity_iSCSI/cbeeecd/DellEMC_Unity_iSCSI/415 : FAILURE in 2m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"26add20fba025e81a9692b3ffe47073ed5d4e604","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-14 12:15:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/800703/1/check/DellEMC_XtremIO_FC/93d3bce/DellEMC_XtremIO_FC/509 : SUCCESS in 1h 15m 28s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/800703/1/check/DellEMC_XtremIO_iSCSI/d491cfc/DellEMC_XtremIO_iSCSI/495 : FAILURE in 20m 22s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"2a88302d36e1bf8a835edbf2eb6bfe0e94ef1dc1","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-07-14 16:16:30.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/800703-1/Lenovo-iSCSI : SUCCESS in 1:14:48\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/800703-1/Lenovo-FC : SUCCESS in 1:46:54","accounts_in_message":[],"_revision_number":1},{"id":"cdfe5c3df926c9f89edf7cdbaa2db82d5a681c62","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-07-14 17:00:05.000000000","message":"Patch Set 1:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/800703/1/fc : SUCCESS in 0h 58m 54s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/800703/1/iscsi : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"15a55257734c644e104ca06fd6bb81ec6275a33e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-17 04:24:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/1/20210717104621/NEC-FC : SUCCESS in 2h 29m 16s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/1/20210717104621/NEC-ISCSI : SUCCESS in 2h 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"da6791911de7eb4994678432f3fbeb080a67794b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-10-27 15:01:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ecc84a4bcf60254c02439ee5f602e4a33749ff54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:31:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf9dcc28c680460ba31aa9b6e419b6df : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af8b8136869b420481b3b4a45dd53cb8 : SUCCESS in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2384a482a47f1625f5f255a54c3716dda0e6751a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-27 16:23:15.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800703/2/iscsi/ : FAILURE in 1h 20m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/800703/2/smb/ : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4085600e46d58cd2333dc66ff61f0bbb5944da3f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 16:27:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/03/800703/2/check/iscsi-driver-primera/b20890c/b20890c.html : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"70d4daba6de09cd703722e17e669727bb3da6ad8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 16:43:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/03/800703/2/check/iscsi-driver-alletra-6k/fff7aed/fff7aed.html : SUCCESS in 1h 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9081a6eaf5c8c7b9980da6dda1fa9410b389f0d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-27 17:13:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_Unity_FC/4ac74c0/DellEMC_Unity_FC/96 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_Unity_iSCSI/cd9553e/DellEMC_Unity_iSCSI/91 : FAILURE in 24m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fbbeac163613a3dcb5effa8cda52b5ce0bb25aae","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 17:26:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_XtremIO_FC/53378aa/DellEMC_XtremIO_FC/56 : SUCCESS in 1h 01m 46s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_XtremIO_iSCSI/0857aac/DellEMC_XtremIO_iSCSI/54 : FAILURE in 29m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"4580fe5d36169311fb435f20adb8ef006764c56d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-27 18:49:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/800703/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f458a60/ : POST_FAILURE in 2h 51m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/800703/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/648b5c1/ : SUCCESS in 1h 50m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ee9c6aa70cc9515f6c05cf3644a0afdcaadec443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 19:04:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c71c88542ffa41ac9eac48e7a63bce28 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40d1a67da4ab4f9991368741b9f24b43 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/365aa846134b401cbdd1048c14555d21 : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d520e689b67440f83cb94a618374583 : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4768d6de541439f8869f304378b01f6 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/51e704844a164af9bfaa84018ed1bdaa : SUCCESS in 52m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/035a6679073946aca12ac92367449032 : SUCCESS in 1h 16m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2277c43fdda545c0aa411f55b1a029be : SUCCESS in 13m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4ae50ce3b481419cb2c0b9353c1c0b72 : SUCCESS in 6m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84ab6a8db4034e38931af2acc7844ac0 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e60c7cbb791d473da439e7d5f69515af : SUCCESS in 7m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9269d1ea2c994b7f8e30d5d0fe9bbb03 : SUCCESS in 7m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a84dc35acc0e4ad29d4798490043fc69 : SUCCESS in 36m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f590037c83d49afbe911e1320001730 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/24177bdc069043608d3343126283d13d : SUCCESS in 1h 51m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/32c9441673c347e38c34921894209b52 : SUCCESS in 1h 28m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d37b9ae7e2fc45e0a9748c5532be76ad : SUCCESS in 1h 48m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d78c72d90b0b4cfca3ab100a3dea4b1b : SUCCESS in 1h 08m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97795ca0fbe04d9d9d8e9e5bdc216b09 : SUCCESS in 1h 07m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/df2f8269be3b42668d0c8928f552e395 : SUCCESS in 38m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98cd37cd4089402da37a5b3c8edd002a : SUCCESS in 1h 24m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce47f9dcb9c04e1a9b2d368ee7c5e986 : SUCCESS in 2h 19m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e985de935e24ff6bb818bbd6aad7a45 : SUCCESS in 55m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea5df58fc3ab4ae0b118b9939dbcaed8 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":2},{"id":"dc50ca9d203c5270aa12e0c21a7265e59ba26b5f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-10-27 19:05:09.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"c7ccefe9989f651485fe2b29d0e3c2730f140005","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-10-27 19:05:20.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"3771ef8d321fdf5530169fc5e8d976c5919afd7a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-27 19:26:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_Unity_FC/04c074d/DellEMC_Unity_FC/101 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_Unity_iSCSI/329a657/DellEMC_Unity_iSCSI/96 : FAILURE in 21m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"72fbbf7bc5d353b7f9d69d0f4679eaeecae49a78","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-10-27 20:21:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/800703/2/upstream-solidfire-check/cinder-solidfire-iSCSI/0127f84/ : SUCCESS in 1h 15m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"c242ca0c28cb9bb10dfed88a3e4b02643d6c37c7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-27 20:25:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/800703/2/check/dsvm-tempest-cisco-zonemanager-job/f2ee5a6/ : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6d3f2c5afc9d390bbe04395723bd499cc3afcb5b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 20:40:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_PowerFlex/ba29e49/DellEMC_PowerFlex/40 : SUCCESS in 1h 35m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9a1d3d3f2b6917d6ba4b18e8c84ded798452fed","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 21:40:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/800703/2/check/iscsi-driver-nimble/006ee14/006ee14.html : SUCCESS in 1h 47m 17s","accounts_in_message":[],"_revision_number":2},{"id":"565dec86b3c25c598e9991d976e721128d13f2f4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 22:01:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_VNX_iSCSI/efbfc39/DellEMC_VNX_iSCSI/37 : FAILURE in 2h 56m 31s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/03/800703/2/check/DellEMC_VNX_FC/b1a35eb/DellEMC_VNX_FC/38 : FAILURE in 2h 25m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"7164ec9cc1dec0356854a24f390ca3fdbfb1e9a5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-28 00:49:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-2/ns5_nfs : SUCCESS in 4h 46m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-2/ns5_iscsi : SUCCESS in 4h 50m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9e6f2ef62bddc073c20424b5b7c8e4e52ae15f2c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-28 12:27:53.000000000","message":"Patch Set 2:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/66/index.html","accounts_in_message":[],"_revision_number":2},{"id":"cff502979651124be5d0cc0d21ac69c889d85c26","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-28 15:11:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/2/20211028211037/NEC-ISCSI : SUCCESS in 2h 54m 25s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/2/20211028211037/NEC-FC : SUCCESS in 3h 01m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4000f42f12fc8c4606b4b1bb29ff678adb5a4f57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-12-15 15:27:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c387e0979fcc87dfc05a42014c101c9e2c9cea7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 15:57:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62fe8cc738f14b43bf80b394c029ed5a : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97b766a4b94f4384842e7f151c33aa0a : SUCCESS in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"993d0212958aa76d8dcbe399961cb2dde236a2f2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-15 16:19:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800703/3/iscsi/ : SUCCESS in 51m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/800703/3/smb/ : SUCCESS in 48m 33s","accounts_in_message":[],"_revision_number":3},{"id":"9a891b330f39b41960fc48a349fd802c44663954","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 17:23:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/03/800703/3/check/iscsi-driver-alletra-6k/5b9e402/5b9e402.html : FAILURE in 1h 01m 46s","accounts_in_message":[],"_revision_number":3},{"id":"10f9658ef8438eb50772e0bf89b35876cc4f14e3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-15 17:27:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/800703/3/check/dsvm-tempest-cisco-zonemanager-job/ba00af8/ : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":3},{"id":"5c458f2cfbf4c3467ade3723a78f8b43f13ddde1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 17:51:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f712f80b4ee465493015042d2a3cb8e : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea7d5f8aef464fec851443fcdb065803 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60add5e498de47aeb7427a7bb5fe5c75 : SUCCESS in 13m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f492ecf78f4f4fae4a44eb55802a4f : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/e90d17ec35254f3da2838482bc349c03 : SUCCESS in 1h 00m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca5b9af92a3c4f82917815d210f28ab1 : SUCCESS in 1h 08m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d28ba7139484e9f9d4625e7ae541351 : SUCCESS in 1h 01m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/94749f8deecb4ac2b8f4562a00bd334d : SUCCESS in 13m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adef9a51797540e9b2854eade6176383 : SUCCESS in 7m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e4e72a8fca45434598e24b0864d7e133 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c7eca1b735f42bdb0a8bea848397c39 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40fb6c312b6e492db2dc73bad2864bfb : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7ed46c117dff45929503638a9344dd8d : SUCCESS in 45m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17c123d94a854c72b52f89d605d0a63e : FAILURE in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2507271f5d74e74baa5f2c3e276b348 : SUCCESS in 1h 20m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7d8fd7a4190b4c09b791d6a6ef8d6801 : SUCCESS in 1h 22m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1abcb6f2a9d4db293a79451892a4113 : SUCCESS in 1h 49m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b88a0b0544d742ae84626dae525dfaff : SUCCESS in 1h 08m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/510caf77b110476b9b4451bf9f8d3a74 : SUCCESS in 1h 09m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/70c4fa59150b45a4bb45729443e77d72 : SUCCESS in 42m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/054f74f2631349f9b6b88ea3c8c43389 : SUCCESS in 1h 19m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6baeb04818a4d4e8a8a7797bb2b5438 : SUCCESS in 2h 15m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3b64e2fa25042348ac659f711f765fb : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":3},{"id":"24f89461d790f08ee64790d294f379088ee52017","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-15 17:55:19.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"a2dfa003323337e812fab63ac672c4879228fca8","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-15 17:58:53.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"5eaa48067c7da828fc46fab0de9679284de9f457","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-12-15 18:02:25.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"868efd41f4f450ac6249ddd505fa3a67a0b58b25","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-15 18:20:46.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/800703/3/check/devstack-plugin-ceph-dsvm-focal-py3/abaa242/ : SUCCESS in 2h 29m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1e92f89041b9bd5c64b6ac19cb03da7cd62718e6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-15 19:15:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/800703/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7162968/ : SUCCESS in 1h 31m 28s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/800703/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ccd1a02/ : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":3},{"id":"73e61e2432f49185b65d1700ec283d0792b04a4c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-12-15 20:21:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-03-800703-3/1639590697 : SUCCESS in 41m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"0e4e57fc4e77cddbf860ad42424290959cc43674","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-15 22:31:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/03/800703/3/ : SUCCESS in 1h 38m 40s","accounts_in_message":[],"_revision_number":3},{"id":"3d9ed53326c0e14305531b5ee98707da9e19d79d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-16 01:40:43.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-3/ns5_nfs : SUCCESS in 4h 53m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-3/ns5_iscsi : SUCCESS in 4h 59m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"4d00240962478352e384393bc1b6e6521844dbe9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-16 06:00:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f3ae040207cc4fd198b36a9a1c109f93 : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":3},{"id":"52f2b89ea432f9eb08a9476758234a2f452ab738","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-16 07:58:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/800703/3/check/iscsi-driver-nimble/8d4d28c/8d4d28c.html : SUCCESS in 1h 51m 41s","accounts_in_message":[],"_revision_number":3},{"id":"375f19891705a1cd116ef32e7357a7778f9e7f4c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-16 11:21:51.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/03/800703/3/check/dsvm-tempest-TOYOU-ACS5000-driver/761e4bd : SUCCESS in 2h 13m 00s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/03/800703/3/check/fc-test/5c03002 : FAILURE in 2h 16m 59s","accounts_in_message":[],"_revision_number":3},{"id":"79288b107af48cc2f1c5478e62d2bed42030ca5f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-12-17 04:41:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/800703-3/PVME-iSCSI : SUCCESS in 58:24.77\n- PVME-FC http://os-logs.tristero.net/dell-ci/800703-3/PVME-FC : SUCCESS in 59:52.86","accounts_in_message":[],"_revision_number":3},{"id":"a2719c2640b374eeabe83b30544b1fd2b994ccc7","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-12-17 14:21:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70601518cd8a0d202c6078b2bcf30bdca014eb3d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-18 08:31:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/3/20211218143602/NEC-FC : SUCCESS in 2h 54m 47s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/3/20211218143602/NEC-ISCSI : SUCCESS in 2h 54m 55s","accounts_in_message":[],"_revision_number":3},{"id":"37f52b9727da029787add1b8d64b2a570d043b9f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-12-22 15:47:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"02c5a36b60c42e13afa307ababa14dae3b15a170","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-11 15:56:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e8aa9f0e441f2ab2c8441925c620d7a09c31ac60","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-01-17 11:37:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db9823a1f27c1f73f460bc92ef9af89b1c1d589e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-01-24 14:53:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6ee24e4f8b8378387855092b82e8b9284c89e280","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-01-24 14:54:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dbb13af24c620fadc0434bf70ea39cbecdc83ef2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-24 15:02:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 1m 13s","accounts_in_message":[],"_revision_number":4},{"id":"7cbf4cb121ec0163a9060aad446e5784ba331c20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 15:16:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d71c9735b7664fb4b4d746802a7c1e26 : SUCCESS in 17m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dde682db01634ed491a9983662dc4afe : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"82d63337008e5b63ce857a91b941581ec10b6a7e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-24 15:26:32.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/800703/800703/4/check-not-vote/ext-cinder-zuul/362e65f : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":4},{"id":"9e653323e472c90fc6ea7f51c74f167e14352310","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-24 16:05:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800703/4/iscsi/ : SUCCESS in 1h 11m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/800703/4/smb/ : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":4},{"id":"5818225c6ef3051d7ddfa04b2d71468871b8acf6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-24 16:34:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/03/800703/4/check/iscsi-driver-alletra-6k/5e22c8f/5e22c8f.html : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":4},{"id":"5e0c8495ac1da8d9350efdcf4043f05699246a85","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-24 16:34:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/800703/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4c6d449/4c6d449.html : FAILURE in 1h 35m 34s","accounts_in_message":[],"_revision_number":4},{"id":"a0b8dbf04c5129ac35acff3252425fc4d65f1d0b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-24 16:47:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/800703/4/check/iscsi-driver-nimble/8c3ccb4/8c3ccb4.html : SUCCESS in 1h 48m 57s","accounts_in_message":[],"_revision_number":4},{"id":"4178dccf8ba3f3a0e0dd5789d397601111f3cf5f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-01-24 16:47:30.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/4/20220125005614/NEC-V-ISCSI : FAILURE in 50m 16s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/4/20220125005614/NEC-V-FC : FAILURE in 51m 13s","accounts_in_message":[],"_revision_number":4},{"id":"28be5a4bb84814b11fa77a24280e9f8d107b1ec3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-24 17:33:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/800703/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3ae9ef9/ : FAILURE in 2h 37m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/800703/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9bf541d/ : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":4},{"id":"c7378aef87ae6b244c5ade28c0e8c38c4bd80867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:38:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a1682697fa4eb5b531d4f11272236d : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff08733a4fab43cbb7475f048405ebce : SUCCESS in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5808eb38a98c4d79a939d42acf0227c2 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8939a1f638fa4ddab23f2afc06519e64 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a2e600dcaf6149b587c29f4c1f28c9d2 : SUCCESS in 55m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad4822f51a3349c99dd4c9800241a616 : SUCCESS in 1h 22m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/852325bc22c4493aa017330fe3b771f3 : SUCCESS in 45m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7680b1864d3141b19f76681054a54530 : SUCCESS in 10m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0b8534aff0014eb7b8d3be9f88f0747a : SUCCESS in 7m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7a61adc83ca4ae7a9a702abf03f5371 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/066c5e0d2bd649c3b71a110381d9de61 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58c8f51e2997498fb3fdc5090596fadf : SUCCESS in 8m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3bc3ef96b7064c6eaaa3f6ace979a524 : SUCCESS in 39m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d99ab0260b824835ba16097459298b75 : FAILURE in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5b8a70691ae48dfaccacccfae87de45 : SUCCESS in 1h 24m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6756606e60fa476eb010c2871b1f16c3 : SUCCESS in 1h 12m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/48fc4d817bc74edd9c7592930a53e4df : SUCCESS in 1h 37m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3189c831872943f095a7a06a7f803af6 : SUCCESS in 1h 10m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3fbc4cea313041289dba18f4f43dcf5b : SUCCESS in 1h 29m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8384b4480cd047f9b5a09fa38f308927 : SUCCESS in 46m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e51d0d13d77d4fa6a01d2298440173f9 : SUCCESS in 1h 24m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b71910363efe4a59a26a103934c3627d : SUCCESS in 2h 38m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65d1645ad272426fbb83b9bb05f82bee : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":4},{"id":"489e536e259711dfc3cf09355e4062bd213489a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-24 17:45:56.000000000","message":"Patch Set 4:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/24_01_2022_17:38:45\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"e0aade61fb8df083ee2ac1da6218ad8f6044db6a","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-24 17:55:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-03-800703-4/1643045918 : SUCCESS in 16m 46s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"9a362fe15f5b2d9f7274dd9495314245aa190993","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-24 18:08:49.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_Unity_FC/8b40b3e/DellEMC_Unity_FC/466 : SUCCESS in 1h 05m 56s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_Unity_iSCSI/b282062/DellEMC_Unity_iSCSI/461 : SUCCESS in 1h 35m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"0320586d5af27b6625fb436bf3984d00f1c26e4c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-24 18:15:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerMAX_FC/7f3bc29/DellEMC_PowerMAX_FC/219 : FAILURE in 36m 54s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerMAX_iSCSI/f7ed0df/DellEMC_PowerMAX_iSCSI/232 : FAILURE in 32m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"003364779c16485e4922c78185a8907c25d42dc3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-24 18:53:02.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_XtremIO_FC/cdc927d/DellEMC_XtremIO_FC/548 : SUCCESS in 59m 11s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_XtremIO_iSCSI/b6658a1/DellEMC_XtremIO_iSCSI/558 : SUCCESS in 1h 25m 22s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"18c9010635d4ace5d95daa8cd724ffea9b0bbed3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-24 19:02:51.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerFlex/7716d02/DellEMC_PowerFlex/306 : FAILURE in 1h 23m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"62aefd30968b2e9a02f1a0607d914544b35b90c9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-24 19:21:46.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerStore_iSCSI/8fef088/DellEMC_PowerStore_iSCSI/275 : SUCCESS in 1h 20m 59s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerStore_FC/049ee1d/DellEMC_PowerStore_FC/278 : SUCCESS in 1h 08m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"c29c9debcf8e244ce2bef852a2830062e6d361c2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-24 20:55:24.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-4/ns5_nfs : SUCCESS in 4h 50m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-800703-4/ns5_iscsi : SUCCESS in 4h 46m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"589dbf0051d5c56771d2877e05467f6c7175d742","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-01-24 21:30:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/800703-4/HPMSA-iSCSI/ : SUCCESS in 51:48.23\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/800703-4/HPMSA-FC/ : SUCCESS in 43:52.92","accounts_in_message":[],"_revision_number":4},{"id":"5d9b56662cb189cbd0868ad4aa2fd19d08c98557","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-24 21:53:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_SC_FC/e329d18/DellEMC_SC_FC/242 : SUCCESS in 47m 34s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_SC_iSCSI/95fee93/DellEMC_SC_iSCSI/240 : FAILURE in 2h 11m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"3b044ec6acc50933f0ab5b5d9f1db79d55da9515","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-01-24 23:43:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/800703-4/Lenovo-iSCSI/ : SUCCESS in 1:03:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/800703-4/Lenovo-FC/ : SUCCESS in 56:27.43","accounts_in_message":[],"_revision_number":4},{"id":"be4433f66bdff619149e51227a977195cdf97d49","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-01-25 00:12:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/800703-4/PVME-iSCSI/ : SUCCESS in 1:04:15\n- PVME-FC https://os-logs.seagate.com/dell-ci/800703-4/PVME-FC/ : SUCCESS in 1:03:25","accounts_in_message":[],"_revision_number":4},{"id":"7c0bce816a836deb3ffd86de14d29fb203ae017a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-25 12:14:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/03/800703/4/ : RETRY_LIMIT in 2m 09s","accounts_in_message":[],"_revision_number":4},{"id":"47fa1b57ba9a8895209992a08c9aad5aa76d9de6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-27 13:06:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/4/20220127154258/NEC-ISCSI : SUCCESS in 3h 22m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/800703/4/20220127154258/NEC-FC : SUCCESS in 6h 23m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7179866116bffe8f269eb21ad86f30dab5976571","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-04 16:24:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a9f8fe08c3cdf987adda1843462d2dd76e9a2c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:29:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0dbb8b59bde7d6fb7158f1bff4ac0bc1dfe4f57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 20:41:20.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a8829da12c435486dda3ba383dcb8d : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b5134c231b94791b0d7f6201bda6c84 : SUCCESS in 12m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dabdb29d91714e68b490d9825a50cc23 : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2cbe1deb884e269557ddd768e31321 : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/aaf8d05c5b764c3ab0062c030173e383 : SUCCESS in 52m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7b237693d214c58bb65d212e650150d : SUCCESS in 1h 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5ade130809043b69ee48defb00a8dac : SUCCESS in 46m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f36115000b2743f28c1a398b699aa046 : SUCCESS in 1h 07m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5955a2811de54c40a9535ccbec441cb4 : TIMED_OUT in 2h 06m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bc0d29b819a44a095108fbf523ed76e : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":4},{"id":"a12ff660d8d7e40a5055c116c375dbe769d7b75d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-02-04 21:51:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dcfff21e7b3f6b96ba1f7d6078c5b1b707d9743d","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-04 23:06:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/f9cdc55/index.html : SUCCESS in 1h 13m 59s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/54fcaa1/index.html : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":4},{"id":"dc7752bf533a660d82acf68e0c609cbc72618e8b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-04 23:16:53.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800703/4/iscsi/ : FAILURE in 1h 24m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/800703/4/smb/ : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":4},{"id":"5cac2eae6c09885a86016d1a2bf2e2e8a01c46f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:17:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3ff66c7a1f14104a996406a6d52681f : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87da25b1765344249b4b47348d822e69 : FAILURE in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02be49fbb538e1a025040ae4fb6b561cc6beb861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:57:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14152be37797457bb8c15a9d5d520f8a : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31013764b0f04a1d9cffd3c95829f030 : SUCCESS in 12m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa3515c18cd455394fae2e33f864daf : SUCCESS in 15m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc808616a87427da6480de2fcba89ac : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/0b9d9ebb74d54235b8104718096c6c6e : SUCCESS in 56m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5827adcdad064cf0a75f6f72fae41945 : SUCCESS in 1h 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0465df26e881400ca288ee9a00693f5a : SUCCESS in 44m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/11b051147e424ed9b678135c5667cfc2 : SUCCESS in 14m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/27c3e1f8adea4ec4a424f5b9880c18e0 : SUCCESS in 7m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/814e6926e3d546d4ae6841806e526a8a : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9908670febe240f8afeb7db0ca5060fa : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac8eac3308f24877bff54f3cf8f8ae32 : SUCCESS in 7m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e63bfa9d9ed64a16a6bf70b576260635 : SUCCESS in 47m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9cdc922512a942c99823698f20fa053c : FAILURE in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd91c74eb1cf4c77af61c41bc5dd39cb : SUCCESS in 1h 17m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08c3a3febb7d4e16927925aef66b6348 : SUCCESS in 1h 20m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e5b3d78be8df482d84c91440b54c07b7 : SUCCESS in 1h 26m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6900fafd18046ffb097edf2c4492507 : SUCCESS in 1h 07m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a0573afebc34e02a2c6204318ab0712 : SUCCESS in 1h 26m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d278b7deccce4a55933ecb27569413be : SUCCESS in 38m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f7cdd516c9214dd880b9b5e8246b24b6 : SUCCESS in 1h 10m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46ad3f4745764e2595c62eb571b405ff : SUCCESS in 1h 59m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3bd9440dee74ddfaa7588e6fd55a4a8 : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":4},{"id":"4dde9f12d89584767755ed640002cca0f45ba4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:58:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"70ec82d53c83ce3dbc7ed5dc2c43db1b2f1f437e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-04 23:59:23.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/800703/4/check/tempest-dsvm-full-focal-py3/ae7823f/ : SUCCESS in 1h 40m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/800703/4/check/devstack-plugin-ceph-dsvm-focal-py3/ff05742/ : SUCCESS in 2h 02m 07s","accounts_in_message":[],"_revision_number":4},{"id":"503d276f83062c83dd2eb4ccbbc797387fc491a4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-05 00:10:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerFlex/60939d6/DellEMC_PowerFlex/368 : FAILURE in 12m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"fa6d1df125f7108b7dbbc3666a9d6e266d3b04e1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-05 00:10:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerStore_iSCSI/35ddc4f/DellEMC_PowerStore_iSCSI/318 : FAILURE in 12m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerStore_FC/ef89f76/DellEMC_PowerStore_FC/335 : FAILURE in 13m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"1f2ca36334f092cc52592bd81206cb92f04777d4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-05 00:13:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_XtremIO_FC/ff0ceae/DellEMC_XtremIO_FC/652 : FAILURE in 15m 19s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_XtremIO_iSCSI/ee58d32/DellEMC_XtremIO_iSCSI/662 : FAILURE in 16m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"7f174bcba85275b7001966cfc2c840265fd84bfd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-05 00:49:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/03/800703/4/check/dsvm-tempest-cisco-zonemanager-job/90f75d1/ : FAILURE in 1h 09m 28s","accounts_in_message":[],"_revision_number":4},{"id":"d0ad6776c93f33bff80853b5b043faa15862beba","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-05 00:51:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerMAX_FC/d2e9605/DellEMC_PowerMAX_FC/280 : FAILURE in 53m 36s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerMAX_iSCSI/fea568f/DellEMC_PowerMAX_iSCSI/293 : FAILURE in 27m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"bdb79ad2dc9d8dcbf6cf669fc27c35264ee32527","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-05 00:51:50.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-check/cinder-cDOT-iSCSI/7c42a4f/ : SUCCESS in 33m 48s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-check/cinder-cDOT-NFS/94a338a/ : FAILURE in 11m 00s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-check/cinder-cDOT-FCP/59619e9/ : SUCCESS in 53m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"137657f906713d71cd2486aed189d133483cca28","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-05 01:42:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_SC_FC/837deb0/DellEMC_SC_FC/302 : FAILURE in 1h 05m 47s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_SC_iSCSI/ffb478a/DellEMC_SC_iSCSI/298 : SUCCESS in 1h 45m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"792bee497574bf1ebe61f0f207da3b5baae5a155","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-05 02:05:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/800703-4/HPMSA-iSCSI/ : SUCCESS in 55:13.96","accounts_in_message":[],"_revision_number":4},{"id":"cad624171750032d812abfb98b703176b739bca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 02:09:20.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9b80442fa741c8896974f2dd879bc0 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8abcb5d42f8d4736b8627d905dff07d4 : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a5aa7f182264420b7c9c20368910870 : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd72efdea5f64a7eb4d65d126a51a0d9 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/bed07243a2704534b3d21b8ab0aed85e : SUCCESS in 46m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2ccf23d08fd4782b4d96bc4cbdd015e : SUCCESS in 1h 01m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98d2c483473046d1bb6a50941009b6d4 : SUCCESS in 37m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae2e2cab94f3420ca161cf9c361d2473 : SUCCESS in 49m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9105fc7b7fc3484481f530852d497827 : TIMED_OUT in 2h 09m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/675444b06e2546e98f318af9be77cee2 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":4},{"id":"0a1616bdf84f855f47df575e1a97796d44708d20","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-05 02:27:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/800703-4/Lenovo-iSCSI/ : SUCCESS in 1:02:28","accounts_in_message":[],"_revision_number":4},{"id":"6d12d25770f5cd26328e6001c87887f004909b4e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-05 02:35:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/800703-4/PVME-iSCSI/ : SUCCESS in 1:06:00","accounts_in_message":[],"_revision_number":4},{"id":"381c1873a6094ea844edd40da73c9cbf8da3ee44","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-05 02:37:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/800703-4/Seagate-iSCSI/ : SUCCESS in 1:07:15","accounts_in_message":[],"_revision_number":4},{"id":"d3d0da698c7fd80632cbd1b57035091eaff470da","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-05 03:25:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-solidfire-check/cinder-solidfire-iSCSI/9f2007e/ : SUCCESS in 3h 26m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"3c1ceeabd9106b50e3cb262b32255634f24453b7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-02-05 16:00:04.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/800703/4/fc : SUCCESS in 1h 28m 12s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/03/800703/4/iscsi : SUCCESS in 1h 22m 55s","accounts_in_message":[],"_revision_number":4},{"id":"11e5195d3e1845640b68c593f10a8f4c6700be20","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-02-07 14:38:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"44ae7bb03ac36a67f00399c4c38539ecb2707a7a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-02-07 14:38:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f7f4723049e4e1bf58d0e7e5caba86f54b5fbaed","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-07 14:44:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/03/800703/4/check/dsvm-tempest-zadara-driver/b524496 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42b3361fefaaeb8f6338178c1c9e7a29fdb1af01","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 14:57:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc045c100463ab2d0daa49002555662023e6b114","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:21:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe9a79ca395f4507b217ef5a5a085323 : FAILURE in 24m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8d2dcc30c0e4905a4b42db248a9ddb3 : FAILURE in 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cf5e5da9203b32a05c38da5790c5911ea497260","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-07 15:51:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/03/800703/4/check/dsvm-tempest-cisco-zonemanager-job/67aa9ee/ : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":4},{"id":"cc733dcec0fde4239984e0d4f2a0c62832b66a09","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-07 15:51:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/e41cdfa/index.html : SUCCESS in 1h 12m 28s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/8b7b17a/index.html : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":4},{"id":"8121894c4e06b1e1470473aac19a06edcfd948a1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-07 15:55:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/800703/4/iscsi/ : SUCCESS in 1h 16m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/800703/4/smb/ : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":4},{"id":"13c17e12f4c6864b3c53cac3c4c9ac3a3b17db4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:42:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2260d4f1d1a64405a960066d13bd8135 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3d254ea10da4188829804d35d3f7041 : SUCCESS in 14m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0dbb60209fa4e729afb2fc856d0253d : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de635b84421b4ba89b000c47903e93fa : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/4aa23d5b6c104a2db0835455d715c7a5 : SUCCESS in 56m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c54913a85da4461f859d781f8640c96a : SUCCESS in 1h 15m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13fc3ae3d94b4e6e97b018db2f90dbad : SUCCESS in 49m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ce5f746900c46a2ac5c7df5b8c6c68f : SUCCESS in 17m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1fe845a6001b4469afce056f175de6fb : SUCCESS in 5m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce1b6067c38f49409a9e4a8592a6a7af : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2acfd335817842b7b5dec8b291234739 : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e7676eef1cd470fab281fbf5771412a : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a87338c3ff4428d968baf9a79f0c0c0 : SUCCESS in 45m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e3cad7c2804c49848590ae837d14ddf1 : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87948cb5895d4a76823907775898bd3e : SUCCESS in 1h 22m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7b3ead244af846a687579d30706a97fb : SUCCESS in 1h 23m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f6cf699a48e24ea09684fceb0c3fb809 : SUCCESS in 2h 25m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbda1ee489bb418fa09f45692d52a25b : SUCCESS in 1h 03m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9654b714177a4f8a92f91eda59ce0422 : SUCCESS in 1h 38m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5ebfb8c40544870abddc4a272b60eb4 : FAILURE in 44m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b460542295714c3680cd0b7935d700f7 : SUCCESS in 1h 18m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9bfef5b2caf748bcb930498a713eef85 : SUCCESS in 2h 56m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3118b5880cb4f5ea24bfd8cdddab6f8 : SUCCESS in 46m 12s","accounts_in_message":[],"_revision_number":4},{"id":"9c27b8c5e5a1d8e88ef27e71326ec49b0e8f242a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:44:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"945c983572d271241b324cb75270ba471c32ee13","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 17:54:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerFlex/d21816d/DellEMC_PowerFlex/375 : FAILURE in 11m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"3749503c884e9c470b7889f4f131c3756fc9090a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-07 17:56:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerStore_iSCSI/8dfbe5d/DellEMC_PowerStore_iSCSI/325 : FAILURE in 13m 23s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerStore_FC/c54857d/DellEMC_PowerStore_FC/342 : FAILURE in 13m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"9d16e7f9a01e7289e21c58eaa3b1b4bac044d65b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 17:58:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_XtremIO_FC/ba17e05/DellEMC_XtremIO_FC/671 : FAILURE in 14m 44s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_XtremIO_iSCSI/f2ce19b/DellEMC_XtremIO_iSCSI/681 : FAILURE in 15m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"dcb5eb34b35fd25ebecacf420881de7fb0bd3e24","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 18:01:32.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/800703/4/check/tempest-dsvm-full-focal-py3/f2cb574/ : SUCCESS in 1h 46m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/800703/4/check/devstack-plugin-ceph-dsvm-focal-py3/092f4a8/ : TIMED_OUT in 3h 12m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2f3d7dd8d3de2cac2ca234de21d23286eb4244f2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 18:35:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerMAX_FC/bef276c/DellEMC_PowerMAX_FC/287 : FAILURE in 52m 38s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_PowerMAX_iSCSI/7fb59cf/DellEMC_PowerMAX_iSCSI/300 : FAILURE in 26m 47s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"f130d0143dec25e427e2a86c22466f894349a643","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-07 19:39:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-check/cinder-cDOT-iSCSI/5f81958/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-check/cinder-cDOT-NFS/649408d/ : SUCCESS in 1h 54m 51s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-check/cinder-cDOT-FCP/a448329/ : SUCCESS in 57m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"c6c731534700aa22085f195925e8288cc66cad3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:52:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3a650c69444cf387d1c4b7841d24dd : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed7dac90f08f4a6aa2dfee80b8e7ae9b : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62da54b4b2804d35ae148f04ac56f404 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c0609dd6ed42d7883a138bb9df5701 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5f6653c65b514e7bb9d36f535c394c36 : SUCCESS in 52m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/571394ac655d448ab21d312eff1efb3e : SUCCESS in 1h 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/368d67652f384247970e91e351b7596a : SUCCESS in 46m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0b95f226d47a4e838003cad2c5af986b : SUCCESS in 1h 02m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/294f8e8fac124f83a5de892b56554046 : SUCCESS in 1h 09m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b9c37110fab49e4824769e348dc1754 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":4},{"id":"3b97f38847ae3c06645c3c633421508c1819ac46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:52:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d6ebeffe1b555368bc83eab4625250052c78a9b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:54:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7078358c7d384fe988e0e974d4a432f6 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f095a41813042c9bc7529f70fd8aa8077d337f26","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-07 20:15:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/800703/4/upstream-solidfire-check/cinder-solidfire-iSCSI/347cd5d/ : SUCCESS in 2h 32m 02s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"5fc61727281c4adc1a04ed9a0546daddba350b56","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-07 21:05:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_SC_FC/9e64dbb/DellEMC_SC_FC/309 : FAILURE in 1h 07m 31s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/03/800703/4/check/DellEMC_SC_iSCSI/9896c6c/DellEMC_SC_iSCSI/305 : SUCCESS in 1h 55m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"36bd8f5c9773caf3c87e4683948bd32ed99ac3cf","revisions":{"9125111c8d6e3e5101c79ff5b6aed542d647ad02":{"kind":"REWORK","_number":1,"created":"2021-07-13 18:02:01.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/800703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/800703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/800703/1"}}},"commit":{"parents":[{"commit":"0570b168e2a444af0d39750638590db0ebcee1a4","subject":"Merge \"Drop lower-constraints jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0570b168e2a444af0d39750638590db0ebcee1a4"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 14:45:37.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 18:00:38.000000000","tz":-240},"subject":"Remove various unused code","message":"Remove various unused code\n\nRemove code that does not appear to be called/used.\n\nChange-Id: Iea1f29ce269107cd6422c40ca2e42ddb7354486b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9125111c8d6e3e5101c79ff5b6aed542d647ad02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9125111c8d6e3e5101c79ff5b6aed542d647ad02"}]},"branch":"refs/heads/master"},"0beaf9f8c24a34baf7b00a4e004afe0ee56575b6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-27 15:01:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/800703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/800703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/800703/2"}}},"commit":{"parents":[{"commit":"f9d6a7243ae8c3bfd70323e2ee1a0f9004fff84d","subject":"Merge \"Pure Storage FlashArray - Version increment for Cinder drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f9d6a7243ae8c3bfd70323e2ee1a0f9004fff84d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 14:45:37.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-10-27 15:01:49.000000000","tz":0},"subject":"Remove various unused code","message":"Remove various unused code\n\nRemove code that does not appear to be called/used.\n\nChange-Id: Iea1f29ce269107cd6422c40ca2e42ddb7354486b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0beaf9f8c24a34baf7b00a4e004afe0ee56575b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0beaf9f8c24a34baf7b00a4e004afe0ee56575b6"}]},"branch":"refs/heads/master","description":"Rebase"},"059200d841fdec635d377e2fefd0b97b171f57e6":{"kind":"REWORK","_number":3,"created":"2021-12-15 15:27:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/800703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/800703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/800703/3"}}},"commit":{"parents":[{"commit":"cf7c2fcc75b5abdc9cb4cac5fb956fb753a4508f","subject":"Merge \"mypy: annotate volume/api.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf7c2fcc75b5abdc9cb4cac5fb956fb753a4508f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 14:45:37.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-12-15 15:27:05.000000000","tz":-300},"subject":"Remove various unused code","message":"Remove various unused code\n\nRemove code that does not appear to be called/used.\n\nChange-Id: Iea1f29ce269107cd6422c40ca2e42ddb7354486b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/059200d841fdec635d377e2fefd0b97b171f57e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/059200d841fdec635d377e2fefd0b97b171f57e6"}]},"branch":"refs/heads/master"},"36bd8f5c9773caf3c87e4683948bd32ed99ac3cf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-24 14:53:40.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/03/800703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/800703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/800703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/800703/4"}}},"commit":{"parents":[{"commit":"5c246e50ec9021edd3ce283822fd7089360c9d20","subject":"Merge \"PowerMax Driver - Re-use existing initiator group/host.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c246e50ec9021edd3ce283822fd7089360c9d20"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-07-13 14:45:37.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-01-24 14:53:40.000000000","tz":0},"subject":"Remove various unused code","message":"Remove various unused code\n\nRemove code that does not appear to be called/used.\n\nChange-Id: Iea1f29ce269107cd6422c40ca2e42ddb7354486b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36bd8f5c9773caf3c87e4683948bd32ed99ac3cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36bd8f5c9773caf3c87e4683948bd32ed99ac3cf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
