)]}'
{"id":"openstack%2Fcinder~802036","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I62fcef3169dcb9f4363a5344af4b2711edfef632","project":"openstack/cinder","branch":"stable/wallaby","topic":"nfs-fix-online-snapshot","hashtags":[],"change_id":"I62fcef3169dcb9f4363a5344af4b2711edfef632","subject":"NFS: Update connection info on online snap create","status":"MERGED","created":"2021-07-23 15:09:33.000000000","updated":"2021-07-30 08:33:50.000000000","submitted":"2021-07-30 08:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802036-nfs-fix-online-snapshot","meta_rev_id":"e4d66e30df5556b72ffd57076508bc23002f3af1","_number":802036,"virtual_id_number":802036,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-30 08:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2021-07-23 20:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-30 07:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-30 07:15:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 15:09:33.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-07-23 15:11:04.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-07-23 15:32:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 16:28:07.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 16:32:27.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-07-23 17:05:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-23 17:09:37.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-07-23 17:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 18:23:44.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-07-23 19:14:12.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 23:37:10.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-24 13:29:01.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-24 19:22:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-30 04:26:40.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-07-30 07:15:36.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"a7a8829a1bd03fa41ffaea7f93a61db6a8732b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-07-23 15:09:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79d8f18127e3fea3e491fd0dab3cd72a29804c76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 15:32:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/126d45e87b314be2b5b5afe6da892a77 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3690b9e12e654412b5c574c48c5e8334 : SUCCESS in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7b94be87faa73af04aa0dfd464190b05d164dea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-23 16:28:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/36/802036/1/check/iscsi-driver-primera/40ffed4/40ffed4.html : FAILURE in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d7adb3b44af20854befe0e0034553e99d58410b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-07-23 16:32:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/802036/1/iscsi/ : SUCCESS in 1h 07m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/802036/1/smb/ : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":1},{"id":"366662b8c59d34bab8c2242e9a28f911875a02a7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-23 17:05:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/802036/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/138a6e5/ : SUCCESS in 1h 54m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/802036/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7b4f914/ : SUCCESS in 1h 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7e0eac8b1f0f3c268c0c5de281be7f939bfbebfe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-07-23 17:09:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/36/802036/1/check/dsvm-tempest-cisco-zonemanager-job/7fcb233/ : POST_FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cc356a66edec9c7969288bfdd3dd6de0ea9b5a5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 17:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e971dff3a8a4983bb53777c714bde7c : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/038dfaad2ee94ab7b1a6f84701080963 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33bfc742f6b54fa2a851311b6b0f7233 : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56876a61a4049199f7f25ce887c9c95 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8a84caa1f5414299ced6fe1b0d4e5b : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/cee4f95c7853482b94b9808c37feec3e : SUCCESS in 1h 13m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ad874a4fab74c0587696c2095b8b3f6 : SUCCESS in 1h 15m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa37f1e1d50400099acd744944ae816 : SUCCESS in 6m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98e9303bb63f46d69cedf4f5d12b6b40 : SUCCESS in 13m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a84014bee68f48639f9757c7c492c702 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/381c72fbb3f04ec0abdd87b0ece2a830 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c1b94906eda4a7cbed97f9f28fd9b61 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8fb8fd4fb97b4e048f022d261e7ebc77 : SUCCESS in 39m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9285c25afb94417da2625a2b79b43cb3 : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9dbf6680e6664b67a3261742915149cc : SUCCESS in 1h 19m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6db1f7cec4c949e8b43fa7f7b4a685b5 : SUCCESS in 1h 20m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0cf391dd43bd4853bf36db2720db52f7 : SUCCESS in 1h 31m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd2eef2107de4a7d96144b22ac57f8dc : SUCCESS in 1h 11m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0a118e023b54332b6522c71447eba00 : SUCCESS in 1h 24m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5d6515dc2f154f1c918e733427dd54da : SUCCESS in 1h 23m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/684f41d2d27e4271b917012b25c3e8c2 : SUCCESS in 2h 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a1245fce3b6476eb7de28d7d39e5e5b : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2f3fa5ede4c1a01225276e96bc10c9964d1c7cd1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-23 18:23:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/36/802036/1/upstream-solidfire-check/cinder-solidfire-iSCSI/abd6a90/ : FAILURE in 51m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"4839f90c2b23056a2782337f5329292bd0cfaca7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-23 19:14:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/36/802036/1/upstream-check/cinder-cDOT-iSCSI/c31a80e/ : FAILURE in 41m 45s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/36/802036/1/upstream-check/cinder-cDOT-NFS/258abaa/ : FAILURE in 42m 54s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/36/802036/1/upstream-check/cinder-cDOT-FCP/5cd862a/ : FAILURE in 1h 42m 02s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"eb9e935d358f2d30f78d2edebc29a2b7590d96aa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-07-23 19:39:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/36/802036/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9c5f8cd/9c5f8cd.html : FAILURE in 50m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f6b69acbad2498272bcd45f53f1cfe29316a6211","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-23 20:37:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e404f1eb6c7552783323c8084ec55ae8053f778f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-23 23:37:10.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-802036-1/ns5_nfs : SUCCESS in 5h 12m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-36-802036-1/ns5_iscsi : SUCCESS in 4h 54m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2deb4e5ebdc29c19cbb866b7a3a1bdafa9a11ac","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-07-24 13:29:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/36/802036/1/ : FAILURE in 46m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3ac5c6cca5a155409e81c3745e03bcb852096034","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-07-24 19:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/802036/1/20210725014242/NEC-FC : SUCCESS in 2h 36m 34s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/802036/1/20210725014242/NEC-ISCSI : SUCCESS in 2h 39m 29s","accounts_in_message":[],"_revision_number":1},{"id":"72144679b3a35a3ca7a3cf1dc7fe1c89e54c6d9d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-07-30 04:26:40.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/36/802036/1/check/dsvm-tempest-TOYOU-ACS5000-driver/b61e781 : FAILURE in 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8d85437184837f06558f1d4a4870b6ef66e8d72d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-07-30 07:15:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged in master (Xena), valid bugfix for wallaby, devstack-plugin-nfs-tempest-full is passing. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"549679882a810aff9c9d9d8d69c61624f4d3a1ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 07:16:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84971b1388505bc11600e79d0d1b9332d180dca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 08:32:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4fbf33e8c684d33af81315b79a5fe9f : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf394e7f96104ae0bb1e57b0acd26313 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98763d2bf651465f863ec3b2d386f998 : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6634149c1f46de845fa9218d35429c : SUCCESS in 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/3792ce15df0545738bd9725e626060b5 : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4788091598104772a69bd3510dc40486 : SUCCESS in 1h 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/860c8ef18a6f4642adee12dea136c670 : SUCCESS in 6m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd64269c42ac41a4ae2de3f41c41cad8 : SUCCESS in 1h 13m 46s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/401e41d90a2a470fbe937ca836fe380b : SUCCESS in 1h 12m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f37f1e4a83743369900591446f9d70c : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5d516095de42b36b070a7cf84930cc0c8ae478ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 08:32:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e4d66e30df5556b72ffd57076508bc23002f3af1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 08:33:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c41c3c95ac467b888a35844334da35 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98afa6bd28844b0587428f0c3d9c01b1 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4960fd597aad8502cad46b38cdb50c6b88f7c63","revisions":{"d4960fd597aad8502cad46b38cdb50c6b88f7c63":{"kind":"REWORK","_number":1,"created":"2021-07-23 15:09:33.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/36/802036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/802036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/802036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/802036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/802036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/802036/1"}}},"commit":{"parents":[{"commit":"77d4aa6a898d25e7fb3ffccd1386b84d3f1ba1d3","subject":"LVM: Retry lvdisplay and lvcreate calls on segfault","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77d4aa6a898d25e7fb3ffccd1386b84d3f1ba1d3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-07-06 11:35:44.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-07-23 15:09:18.000000000","tz":120},"subject":"NFS: Update connection info on online snap create","message":"NFS: Update connection info on online snap create\n\nThe NFS snapshot creation for an attached volume requires interaction\nbetween Nova and Cinder, and a new qcow2 file is used after the\nattachment completes.\n\nThis means that the connection properties stored in the Attachment is no\nlonger valid, as it is pointing to the old qcow2 file, so if Nova tries\nto use that attachment it will start writing on the old qcow2 file.\n\nA flow showing this issue is:\n\n- Attach NFS volume\n- Create snapshot\n- Hard reboot\n\nAfter that the VM will start using the base image, breaking the qcow2\nchain.\n\nIf we delete the snapshot in the meantime, then the VM will fail to\nreboot.\n\nThis patch fixes this inconsistency by updating the connection info\nfield inside the remotefs driver.\n\nWe usually prefer that drivers don\u0027t to touch the DB, directly or\nindirectly (using OVOs), but in this case we are using OVOs methods\ninstead of the usual model update on the volume manager because there\nare cases in the driver where a snapshot is created (cloning via\nsnapshot) and we have to update the attachment without the manager, as\nit is unaware that a temporary snapshot is being created.\n\nBesides that main reason there are other less critical reasons to do the\nattachment update in the driver:\n\n- Smaller code change\n- Easier to backport\n- Limit change impact on other areas (better for backport)\n- The snapshot_create model update code in the manager does not support\n  updating attachments.\n- There are cases in the cinder volume manager where the model update\n  values returned by snapshot_create are not being applied.\n\nSnapshot deletion belonging to in-use volumes are not affected by this\nissue because we only do block commit when the snapshot file we are\ndeleting is not the active file.  In _delete_snapshot_online:\n\n        if utils.paths_normcase_equal(info[\u0027active_file\u0027],\n                                      info[\u0027snapshot_file\u0027]):\n\nCloses-Bug: #1860913\nChange-Id: I62fcef3169dcb9f4363a5344af4b2711edfef632\n(cherry picked from commit 25eb0a7d76922e3a1a289d26c36b96a91c4059db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4960fd597aad8502cad46b38cdb50c6b88f7c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4960fd597aad8502cad46b38cdb50c6b88f7c63"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
