)]}'
{"id":"openstack%2Fcinder~805144","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I0c0cb5308246a3dce736eade67b40be063aa78bb","project":"openstack/cinder","branch":"stable/wallaby","hashtags":[],"change_id":"I0c0cb5308246a3dce736eade67b40be063aa78bb","subject":"LVM: Retry lvextend commands on code 139","status":"MERGED","created":"2021-08-19 12:24:15.000000000","updated":"2021-08-24 17:03:04.000000000","submitted":"2021-08-24 17:01:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805144","cherry_pick_of_change":805026,"cherry_pick_of_patch_set":2,"meta_rev_id":"d717957dc38b5d1c1962fa293cd6abb75847af8f","_number":805144,"virtual_id_number":805144,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 17:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-08-20 14:40:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-20 14:37:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2021-08-20 14:10:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":2,"date":"2021-08-24 15:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":1,"date":"2021-08-24 15:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-08-20 14:40:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 12:48:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:08:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:23:55.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-08-19 13:52:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 14:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 15:00:43.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-08-19 15:20:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 15:20:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-19 15:58:51.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-08-20 00:26:56.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-08-20 14:10:43.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-08-20 14:37:06.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-08-20 14:40:58.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-24 15:04:31.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"c53653ec4f4a5ca54b5705e6334cc4d243c89894","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-19 12:24:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1bb31caa736a17b08facd3b3e3b52121aea1d864","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 12:48:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9e82b3a22704a499bf8ff94e1265c31 : SUCCESS in 14m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d05266fbe15f490288c182ee1bd90b2d : SUCCESS in 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38efed5be7b925d3d644e192f191673a2cf27d7f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-19 13:08:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/44/805144/1/check/iscsi-driver-nimble/7c35288/7c35288.html : FAILURE in 43m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0f8ebf16954d07fe16f0f0097351798f50699750","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-19 13:23:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/805144/1/iscsi/ : SUCCESS in 58m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/805144/1/smb/ : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a0eaccf08b66439ae0599bbb5aca3c8b62f4eabf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-19 13:28:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/44/805144/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/21450fe/21450fe.html : FAILURE in 33m 35s","accounts_in_message":[],"_revision_number":1},{"id":"24f2b24b9c4beeb22abe0f5b8e487c73d8aa65ab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-19 13:30:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/44/805144/1/check/iscsi-driver-primera/30b8094/30b8094.html : FAILURE in 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0e510ff28e3d33efc0de842c2e0e6f66ff3b34c5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-19 13:52:26.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-44-805144-1/ns5_nfs : FAILURE in 1h 12m 30s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-44-805144-1/ns5_iscsi : FAILURE in 1h 12m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a435e3d3874587840355d61bf44b6ca4e99ca85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:43:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc640e704e64d4086980b7f17c012e1 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2ddc82248804b62b1ca95684c0a7716 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c2daf6d1ee84a7f899b04e633f5497d : SUCCESS in 9m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b03c269f1c84b068c4e29b729fee291 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccab51cda10477fb3a4a9f110c2c2ea : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/1fd7f69deeae4ec1867f0d64e90f39b2 : SUCCESS in 1h 03m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46d2285442c54d57b139c8ef8fa7bc91 : SUCCESS in 1h 08m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8b61304cbd3f4d2d9f4b7e70688c5b00 : SUCCESS in 13m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f0309c58b66449eb7f56bdad9615b52 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45c23f4a4a4d4c8b8f2995e136fcbd34 : SUCCESS in 6m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1ff6861ce9b480b87da8af8bee5a9d0 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af376afb0ea94ad9ba3c3ea574985bb2 : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d01ac3e70f5942609e811523bb3f6292 : FAILURE in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e8611ddc4c6483b934dd8272340f14a : SUCCESS in 1h 15m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4bafad94503f40ed9b4c4ac9b19f2e2a : SUCCESS in 1h 39m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cfd03e5ed2a14c1c922bfd88c3410db9 : SUCCESS in 1h 30m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a70206800e74dc9984e27530cab6684 : SUCCESS in 1h 12m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/69cfd3fe18b54264b12c3ca2fee2698c : SUCCESS in 1h 18m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b629a9e9aec494dae7aee978a350a1e : SUCCESS in 1h 25m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efb557c2ff9e4c308352e7e83ee02047 : SUCCESS in 2h 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4297dda40b44e6d8baccaf7b8a2f388 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":1},{"id":"56f819c95675bb046117e894a678e6bab6d6d336","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-19 15:00:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/44/805144/1/check/dsvm-tempest-cisco-zonemanager-job/f4647cb/ : FAILURE in 1h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"aae8678a04c4a0f78f06a5c182309356a200bf51","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-19 15:20:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/44/805144/1/upstream-check/cinder-cDOT-iSCSI/151a31c/ : FAILURE in 34m 32s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/44/805144/1/upstream-check/cinder-cDOT-NFS/5fa8d35/ : FAILURE in 34m 13s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/44/805144/1/upstream-check/cinder-cDOT-FCP/edc6806/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4dfcdadb48fdbd60c3027d05cbecf4fffee32dda","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-19 15:20:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/805144/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/909b8c9/ : SUCCESS in 2h 39m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/805144/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/506566c/ : SUCCESS in 1h 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8aa659b9beaf78fe0e3e91218dd24ddb5928f15e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-19 15:58:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/44/805144/1/upstream-solidfire-check/cinder-solidfire-iSCSI/7bbfc24/ : FAILURE in 1h 13m 37s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"9a1953540ced46bfc58c6a6ed857ac78836413b8","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-20 00:26:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/44/805144/1/check/dsvm-tempest-TOYOU-ACS5000-driver/ae28b6e : SUCCESS in 1h 31m 46s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/44/805144/1/check/fc-test/551c77f : SUCCESS in 1h 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"97bf6f9edbc7a01f0a724e9a9a78895e800f7960","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-08-20 14:10:43.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks like a clean backport","accounts_in_message":[],"_revision_number":1},{"id":"560729225a6cfcea03d7622209554aad3fd060e6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-20 14:37:06.000000000","message":"Patch Set 1: Code-Review+1\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"8c0cf66ebb35357798407b1102ec9f225078a37c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-20 14:40:58.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nAgree with Sofia, we should backport this one.","accounts_in_message":[],"_revision_number":1},{"id":"f7aa9b5c7ff6017c0a4a853eac0abedb9e2029be","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-08-24 15:04:31.000000000","message":"Patch Set 1: Code-Review+2\n\nYes, we do want to get these backported.","accounts_in_message":[],"_revision_number":1},{"id":"b5f1bfd3366178bb786fe8cb36d3b6ba4a107244","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-08-24 15:04:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c4859c60ad3e477f146008ca9e6ff13aa9b8c99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:05:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61f96a7fe2b2b0329d91a79a4aec4121903bf5e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:01:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13fde473ede4adfbcb546071e74e767 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76539cdf22134ab28980e2b5693951e8 : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed547f441c504041a9c271c270942229 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d33e56261c4cef994c74b6ea9d0269 : SUCCESS in 14m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/03656db7b6ba4318bd0b8f22162ea1fd : SUCCESS in 58m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3e3c404cf8d46b3a7b2ba84ee4b4241 : SUCCESS in 1h 10m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5494487736584b0fa3b048ab14c0b66b : SUCCESS in 1h 12m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ddfa8e2df7e24dc9942b3fca89a5dff1 : SUCCESS in 1h 19m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/913f28295f474342bc44f9888159906f : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"890caba686f9686be40c3fab5904df0f919c1e01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:01:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d717957dc38b5d1c1962fa293cd6abb75847af8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:03:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63001570343848e5a168ca08286a03b4 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2425f3ef590524c357622748df79d87edebd17f3","revisions":{"2425f3ef590524c357622748df79d87edebd17f3":{"kind":"REWORK","_number":1,"created":"2021-08-19 12:24:15.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/44/805144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/805144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/805144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/805144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/805144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/805144/1"}}},"commit":{"parents":[{"commit":"96d3dd39b123ef7afe96365e74366836d96b6e8c","subject":"Merge \"[Pure Storage] Resolve missing provider_id issue (PowerVC)\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96d3dd39b123ef7afe96365e74366836d96b6e8c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-18 13:19:58.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-19 12:24:15.000000000","tz":0},"subject":"LVM: Retry lvextend commands on code 139","message":"LVM: Retry lvextend commands on code 139\n\nRetry lvextend commands upon segfault, similar to other\nLVM calls.  This affects the volume extend path.\n\nChange-Id: I0c0cb5308246a3dce736eade67b40be063aa78bb\nRelated-Bug: #1901783\nRelated-Bug: #1932188\nCloses-Bug: #1940436\n(cherry picked from commit c4b8956763c2759bd2b9fe511efac5b3edfb226d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2425f3ef590524c357622748df79d87edebd17f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2425f3ef590524c357622748df79d87edebd17f3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
