)]}'
{"id":"openstack%2Fcinder~805283","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~Ieb19d5e3f58ae2343b6b145772fec33cb7517ab5","project":"openstack/cinder","branch":"stable/wallaby","topic":"bug/1842088-stable/wallaby","hashtags":[],"change_id":"Ieb19d5e3f58ae2343b6b145772fec33cb7517ab5","subject":"Allow removing NFS snapshots in error status","status":"MERGED","created":"2021-08-20 18:21:33.000000000","updated":"2021-08-31 10:24:37.000000000","submitted":"2021-08-31 10:22:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805283-bug/1842088-stable/wallaby","cherry_pick_of_change":679138,"cherry_pick_of_patch_set":12,"meta_rev_id":"c9fd61d7f6ec58ff6aaccb685e099b548044e184","_number":805283,"virtual_id_number":805283,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 10:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-24 15:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-08-31 08:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-08-31 08:50:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 18:43:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 19:07:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 19:23:02.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 19:29:04.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-08-20 19:48:40.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-08-20 20:20:52.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-20 21:24:50.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-08-20 22:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-21 00:43:59.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-21 01:01:58.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-08-23 07:33:32.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 15:03:45.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-08-29 22:39:47.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 08:50:42.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"5bb3f9299c947f9382e9907e2441a22421efffdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-20 18:21:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"afceaf7a13a251d17c60a4760799238b64c8c116","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 18:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7cda4524a954cb0994eea7bb2e84e04 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4094283152f84f28a027b5ba679461a5 : SUCCESS in 19m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85b858b4cab85fae560a525fb56862b82d1ed68b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-20 19:07:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/83/805283/1/check/iscsi-driver-nimble/7fa9a8a/7fa9a8a.html : FAILURE in 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a80a05ef9c71c223b5caaab7ee59148983d8b2ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-20 19:13:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/83/805283/1/check/iscsi-driver-primera/bb029af/bb029af.html : FAILURE in 34m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e76f21fcd78e72f4946a39dddc38771c2275ead3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-20 19:14:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/83/805283/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/394b34c/394b34c.html : FAILURE in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7b5e9d083ff6300146c44b89242a3f794d647308","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-20 19:23:02.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-83-805283-1/ns5_nfs : FAILURE in 1h 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-83-805283-1/ns5_iscsi : FAILURE in 58m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"1e80dcb4c0664d8f257af1bc3888b7cd52691733","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-20 19:29:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/805283/1/iscsi/ : SUCCESS in 1h 06m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/805283/1/smb/ : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"03f410e72c2e16d121ca2860b2e79e229491992f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-20 19:48:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/83/805283/1/check/dsvm-tempest-cisco-zonemanager-job/941267a/ : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2279854238b529af25dc2c960539fdea5cb18d24","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-20 20:20:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/83/805283/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/dd34e8b/ : FAILURE in 31m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/83/805283/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7118f09/ : SUCCESS in 1h 57m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7a80b55188ab051b77cbce73435908f8198583e4","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-08-20 21:24:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/83/805283/1/check/dsvm-tempest-TOYOU-ACS5000-driver/466001f : SUCCESS in 1h 32m 36s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/83/805283/1/check/fc-test/c5c7efc : SUCCESS in 1h 43m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e5689a1e0f7eab2806b96ff1bbd1a900009650e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 22:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77d9622f51b4cc2a6cbe8029de50d02 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6628c48bf5a49308a32c760928ed65e : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d2cce102a6f42f384cf4fc2e9c136b5 : SUCCESS in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d824fe42d1344ea8b257bbc6f12b76e8 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8874437285404adba724750da53202e1 : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/0e98348f759d4c8b91a3cc9cf8a77627 : SUCCESS in 58m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8fe6fbaed4e4150837bd1423742c69a : SUCCESS in 1h 15m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/62c45782b02e40958e53029f27bb8a3b : SUCCESS in 13m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad775199de9d47c1a48d577e212c1e05 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe0dce79658f4c75a59a8544227463a6 : SUCCESS in 8m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/032e80ab349347ecb6668e790ee8bd5c : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a9f49f77d2a4606947b8f53c0647972 : SUCCESS in 43m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6845a1afd6284510b5f204e5f008ef18 : FAILURE in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/66516d595cec49e5b9274b68e7fca79e : SUCCESS in 1h 22m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/755235f632174b6cb3c529aa88bdc9f2 : SUCCESS in 1h 37m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f61f139c991942f1b35a8ef7a4615613 : SUCCESS in 2h 06m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d6decd23a3549379d58fbdfdda8666e : SUCCESS in 1h 14m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7bef30c862d4168a3db09de93416440 : SUCCESS in 1h 28m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f39a4c6224994922a5a95245c1038bf7 : SUCCESS in 1h 23m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b029e15ee51e4a63a94798519884bf17 : SUCCESS in 2h 22m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5af779d7e91f4cc9b4689ab243c3d4a8 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"13dce2c0d837bbb287aaa5e71ef8999c2b5e8c21","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-21 00:43:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/83/805283/1/upstream-check/cinder-cDOT-iSCSI/35ae399/ : FAILURE in 2h 02m 09s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/83/805283/1/upstream-check/cinder-cDOT-NFS/6f393b4/ : FAILURE in 2h 01m 38s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/83/805283/1/upstream-check/cinder-cDOT-FCP/ce28e65/ : FAILURE in 57m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c04a9d8d783d398276d9db59c551fd02b94a317","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-21 01:01:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/83/805283/1/upstream-solidfire-check/cinder-solidfire-iSCSI/69c63c3/ : FAILURE in 2h 19m 50s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"51878b15d0c5685e0a1d0023917a365a649ecab7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-08-23 07:33:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/83/805283/1/ : SUCCESS in 2h 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"76c4d87eaaf493b05be0a2650c551d68ed2d1573","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-08-24 15:03:45.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is a good backport.","accounts_in_message":[],"_revision_number":1},{"id":"bac9f975d510c52e6e4d157fe8403eef2dd0cd09","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-08-29 22:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/805283/1/20210830023647/NEC-ISCSI : SUCCESS in 2h 34m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/805283/1/20210830023647/NEC-FC : SUCCESS in 5h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a7808cb5e8b07c824f274fcf807045127d5541de","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-31 08:50:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged in master (Xena), bugfix for nfs driver, looks appropriate to backport.","accounts_in_message":[],"_revision_number":1},{"id":"dbdfb59039c21b3544f8156da4e6d53f6feecd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 08:51:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bca524f08d01487a4e4279d17a1766ede001ace9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:22:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3120914cfb4d4db0dbdfd413bb0cd1 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47cd5cb8b86a4d19984caf27027bcfac : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/024839a524f74a5eab3a81719e5d23ba : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1154b05e862d433eb8a3d2d3ed382ad6 : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/10642656ea2c434fa01fa9f1e4e4bad6 : SUCCESS in 58m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d59ee77683964d41947f7bb1c0fbcfff : SUCCESS in 1h 13m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee18b0e27f304a209bb3b26af1c1c8d7 : SUCCESS in 1h 15m 40s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1eb86731deb94dfe945d0d0fb610bb14 : SUCCESS in 1h 29m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0285e9edb25c4a7eb1eb9b692881f3c5 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"564d401256d1dd16e268ca986ef25f84d10d0ac9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:22:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9fd61d7f6ec58ff6aaccb685e099b548044e184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:24:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ebc798454043298ad5827cbd774a49 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a4f3ac53a554ecaa18cbd0a49c0606dff4e1137","revisions":{"5a4f3ac53a554ecaa18cbd0a49c0606dff4e1137":{"kind":"REWORK","_number":1,"created":"2021-08-20 18:21:33.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/83/805283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/805283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/805283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/805283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/805283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/805283/1"}}},"commit":{"parents":[{"commit":"1732de19e6c41e18922d78ab590188c4e1ad8465","subject":"Merge \"Add libcgroup related packages in bindep.txt\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1732de19e6c41e18922d78ab590188c4e1ad8465"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2019-08-28 20:38:50.000000000","tz":-180},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-08-20 18:21:33.000000000","tz":0},"subject":"Allow removing NFS snapshots in error status","message":"Allow removing NFS snapshots in error status\n\nThe NFS driver doesn\u0027t allow to delete snapshots in error state when\nsnapshot support is disable.\n\nI\u0027m facing some scenarios where snapshots are enabled then a snapshot\nis created and after that then snapshots are disabled. This results\nin error state snapshots that can\u0027t be deleted.  Another scenario is\n- with nfs snapshot support disable from the beginning- the API layer\nallows the snapshot to be created but the NFS driver sets the status\nto \"error\" leaving the snapshot in the DB with error state.\nBecause of that, I have snapshots in error state that I\u0027m not able\nto delete.\n\nThe purpose of this fix allow deleting snapshots in error state.\nIt makes sense to block creating snapshots when snapshot support is\nFalse. However, when deleting snapshot we don\u0027t want to block\nattempts to delete a broken snapshot DB entry, which is necessary to\nclean up a failed snapshot create.\n\nCloses-Bug: #1842088\nChange-Id: Ieb19d5e3f58ae2343b6b145772fec33cb7517ab5\n(cherry picked from commit 5714bdb8447a38440585116f784a322143711959)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a4f3ac53a554ecaa18cbd0a49c0606dff4e1137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a4f3ac53a554ecaa18cbd0a49c0606dff4e1137"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
