)]}'
{"id":"openstack%2Fcinder~805773","triplet_id":"openstack%2Fcinder~master~I4bae6b684d7f82c171936c455fc0a4b0b9e8dd2b","project":"openstack/cinder","branch":"master","topic":"fake-policies","hashtags":[],"change_id":"I4bae6b684d7f82c171936c455fc0a4b0b9e8dd2b","subject":"DNM: Fake policies - AA","status":"NEW","created":"2021-08-24 02:04:24.000000000","updated":"2021-08-31 05:17:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31832052d7ca32d4a902262dda8258621ea8c660","_number":805773,"virtual_id_number":805773,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-24 03:00:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 02:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 02:32:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 03:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 03:08:52.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 03:11:59.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-24 04:52:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 05:57:39.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 06:31:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 07:22:20.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-08-24 07:51:09.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-08-24 19:04:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 10:54:44.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 01:29:07.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4fe952cd0cc0909de53f6ca2792f1696d70e0b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-24 02:04:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a311b50cdd2e9412cbc33a72d6b96373f79fd428","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-24 02:07:51.000000000","message":"Patch Set 1:\n\nsample config file for this patch: https://paste.opendev.org/show/808270/","accounts_in_message":[],"_revision_number":1},{"id":"35c8ee7ae7eeba9f7720d0acd9e7a51c1eb91f29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 02:28:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3daccbbd0b749cbb33e0bae44d90327 : FAILURE in 14m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d8eb5393e2c49b89c07597786e1231c : FAILURE in 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"957c95ab0741a0473cde76f06fe6541f812ffe4f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-24 02:32:54.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/22f9b51043104835adb75ce3bacd701a : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/92e66d24c4c646e5916e9efbe3fa31de : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c92979a14b56c04a60ddddebeb58e74621d37f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 03:00:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0614280befa4c389a5da909ca1c4965 : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4fc964ff3e1478098c69653cb4d9279 : FAILURE in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bc30068606b45cab62015d29a841e82 : FAILURE in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9052b9f92f5477eb6a3e087af5c047c : FAILURE in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933f7f034f474a18977e674b3f49b7a0 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/e67cf8c228c54535b3a7a8f6f9068bea : FAILURE in 44m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/edce9a3b4e294725a2571c983f7686f7 : FAILURE in 19m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0b6058ff8bbd44dc8a4a60848d413a4e : FAILURE in 10m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47b3d27fcb4e43088623f8a490784f77 : SUCCESS in 6m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bdcd0fbb36774c23b4b0666b2ec50822 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d9d7333e1c74ec38aa916ca955d65f0 : FAILURE in 8m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a456889636744c7cb56436231c964dc3 : FAILURE in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12556f71d2e943fe92209d6158591288 : FAILURE in 20m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f25b128f82904c8f960ed73699a2f05c : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04d5ff9599b240589380c5486204ce0a : FAILURE in 23m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/753efebfe2fb43d8bd2ac5e19c8b2d55 : FAILURE in 33m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c51936d5b0064415a07237b0b2cdca15 : FAILURE in 22m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/87013d09253145a8945e5d5ef39f0497 : FAILURE in 53m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0242cfb87f6f4e7ea535cf2ea3021545 : FAILURE in 28m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bdc599af6a1d493a85b72a3e8b836ebb : FAILURE in 31m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc5f4fd6337c439f8f2dd983ed588719 : FAILURE in 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a5817d43d1740bda637581175442d28 : FAILURE in 22m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e1fad7bb28c41fc8302451a9a5ffa7e : POST_FAILURE in 21m 26s","accounts_in_message":[],"_revision_number":1},{"id":"96a8f7e96268e982d44efc8f5c453c1a8c8cf6c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-24 03:08:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/73/805773/1/check/cinder-storpool-tempest/1100183/ : RETRY_LIMIT in 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"cc78ccf91769494e4156bc1ae8d90b020b9ee3d2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-24 03:11:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/805773/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a92a4f3/ : FAILURE in 21m 03s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/805773/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bd6cd7e/ : FAILURE in 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9085739d1139a4fe206b587856bd3b5305972813","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-24 04:52:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/73/805773/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cc6a62d/cc6a62d.html : FAILURE in 52m 43s","accounts_in_message":[],"_revision_number":1},{"id":"20383e6b5c70d2dd65fba6b1763c946eeeb7e091","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-24 05:09:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/73/805773/1/check/iscsi-driver-primera/6969481/6969481.html : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"db03fae0be20a065862b4654f9910ac34cd7e836","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-24 05:42:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/73/805773/1/check/iscsi-driver-nimble/5cd4453/5cd4453.html : FAILURE in 38m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2e2fe54a7195a521340e9495ea87679461fd7571","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-24 05:57:39.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-805773-1/ns5_nfs : FAILURE in 54m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-805773-1/ns5_iscsi : FAILURE in 55m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f2c98c38ad95e9c06a0022f1e6e84299a59387d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-24 06:31:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/805773/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0d0d8001604b6b7473a4e0661f1d506e8ffe214b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-24 07:22:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/73/805773/1/check/dsvm-tempest-cisco-zonemanager-job/339c1e2/ : FAILURE in 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f1b32b74153cc5b4ce329a84a43203d089523443","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-24 07:51:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/805773/1/iscsi/ : FAILURE in 2h 55m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/805773/1/smb/ : FAILURE in 4h 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a67b0e5097fe4f591a2b0b7bded9f01fb0d4a962","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-24 08:34:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/805773/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c3ebd235229381a6ea55753ae374ce20067d7468","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-24 19:04:54.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/805773/805773/1/check-not-vote/ext-cinder-zuul/6c872db : FAILURE in 23m 32s","accounts_in_message":[],"_revision_number":1},{"id":"beb6357255e9620423d3b16bc8368c5168d219be","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-25 10:54:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/73/805773/1/check/DellEMC_PowerMAX_FC/faec168/DellEMC_PowerMAX_FC/393 : FAILURE in 9m 03s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/73/805773/1/check/DellEMC_PowerMAX_iSCSI/b7f0b3f/DellEMC_PowerMAX_iSCSI/401 : FAILURE in 8m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"82b70d0904beafeb535209b025a8cdf65c580fef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-25 14:23:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/805773/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f8a9c15b7280de376ea3713b159f39a45565335d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-08-31 01:29:07.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/805773/1/20210831091131/NEC-FC : FAILURE in 1h 07m 54s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/805773/1/20210831091131/NEC-ISCSI : FAILURE in 1h 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"31832052d7ca32d4a902262dda8258621ea8c660","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-31 05:17:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/805773/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9dd382ffd40b875f46e624efe491f8fc36d015e9","revisions":{"9dd382ffd40b875f46e624efe491f8fc36d015e9":{"kind":"REWORK","_number":1,"created":"2021-08-24 02:04:24.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/73/805773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/805773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/805773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/805773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/805773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/805773/1"}}},"commit":{"parents":[{"commit":"86f5e6e3eeda35955c95dbc354aa8678e3b80966","subject":"DNM: Fake policies - yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86f5e6e3eeda35955c95dbc354aa8678e3b80966"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-08-24 01:38:25.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-08-24 01:38:25.000000000","tz":-240},"subject":"DNM: Fake policies - AA","message":"DNM: Fake policies - AA\n\nChange-Id: I4bae6b684d7f82c171936c455fc0a4b0b9e8dd2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9dd382ffd40b875f46e624efe491f8fc36d015e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9dd382ffd40b875f46e624efe491f8fc36d015e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
