)]}'
{"id":"openstack%2Fcinder~805825","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~I0c0cb5308246a3dce736eade67b40be063aa78bb","project":"openstack/cinder","branch":"stable/victoria","hashtags":[],"change_id":"I0c0cb5308246a3dce736eade67b40be063aa78bb","subject":"LVM: Retry lvextend commands on code 139","status":"MERGED","created":"2021-08-24 17:13:51.000000000","updated":"2021-09-17 00:07:49.000000000","submitted":"2021-09-15 22:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805825","cherry_pick_of_change":805144,"cherry_pick_of_patch_set":1,"meta_rev_id":"5cf9e8a427de945eb7a996d6b2c01f1832eb430d","_number":805825,"virtual_id_number":805825,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 22:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-17 00:07:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-09-17 00:07:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-15 18:38:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 17:37:06.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 19:10:25.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-08-24 19:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 21:11:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 01:25:54.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 01:27:35.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-25 11:44:48.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-09-02 15:00:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-10 12:26:58.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-09-15 07:43:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"8f1c13a854d07c523a4035a67199e243ccb0e4f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-24 17:13:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"8c419c6d9fc4ace87396753ef13d68d03722293c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-24 17:37:06.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c315acbdb4cb4f1f9acfc5193beb3ba6 : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/6af8bdc56508445c95ec2a96eec4a963 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"78da05b8b01600cf315be0376f68ee4ae97b1aac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-24 19:10:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/805825/1/iscsi/ : SUCCESS in 1h 02m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/805825/1/smb/ : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b42650bbd5ea427df858319dd7bc4dea4a8adcc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:36:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18165a09a524c91b86fc900d7e1bdd4 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbad4a8b79da4c83a67313764f83950c : FAILURE in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b42ca11cee844eb688880633ccd9da68 : FAILURE in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0224d6c887242f685cfeefcdf55e475 : SUCCESS in 15m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/185d32dddc0441df9d3f28a9c192f8a6 : SUCCESS in 58m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cd52984acc9415ba9d0cc6792e3280a : FAILURE in 1h 02m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2bc006e3b8d94ae79a0ac39d47748caf : FAILURE in 10m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86b9d0e6632442f4bad9302b329ec467 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc49814eef8840f8bd9e0bb94f134ab8 : SUCCESS in 6m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/539a669b5cc6497790ab78ab38be4849 : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e3aa7ae17a94344be25531e1f65e6df : FAILURE in 1h 02m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d0f97754d1044ae484bfbb7d35067000 : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8f44211aac74af1affe9f10075faa71 : SUCCESS in 1h 15m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9043b2872e64bafbf5f11253898b5cc : FAILURE in 1h 52m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73cb38efb907436fb571fe7e478caa32 : FAILURE in 1h 37m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/36cd2e8e87214888a660b3b9a6d05d7f : SUCCESS in 1h 09m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9de010776ad04047a8af01b15ff0bcbf : FAILURE in 1h 17m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43e9a0d92dca4353b7528df0e89e3ed5 : SUCCESS in 1h 24m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0ac9f7c2caab42bd80cd33bffb4ebf0b : SUCCESS in 2h 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23e1e078347c49159c235118570c327c : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"032926486cb624dfdbe6ca7f68ac8147a886400b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-24 21:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/25/805825/1/check/iscsi-driver-primera/2a75729/2a75729.html : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"83c49e0bb3d512bd6d06f776628e2792f556a46c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-24 21:59:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/805825/1/check/iscsi-driver-nimble/687088f/687088f.html : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"07370816f69ecb6c0381b0ea00ad3d451074f0dd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-25 01:25:54.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-1/ns5_nfs : FAILURE in 1h 19m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-1/ns5_iscsi : FAILURE in 1h 21m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"16dda956503691d22814bf509f949a76bf541cfe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-25 01:27:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/25/805825/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/22f4dc8/ : SUCCESS in 2h 33m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/25/805825/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0db9e1c/ : FAILURE in 2h 46m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1ada74eb8fdab770c704e436565a8243408cce23","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-25 11:44:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/25/805825/1/check/dsvm-tempest-cisco-zonemanager-job/025754b/ : SUCCESS in 1h 18m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3df7a299629d9cbd3f6a9ff08e5d1753b2fea2b9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-25 14:40:58.000000000","message":"Patch Set 1: Code-Review-1\n\nunit test failure looks legitimate","accounts_in_message":[],"_revision_number":1},{"id":"3073adecfaa45ecda775b3c39621892fad34739c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-25 14:56:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"25119f57e6b926b4385df9481798c468bf226c74","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-08-25 15:23:56.000000000","message":"Removed Code-Review-1 by Eric Harney \u003ceharney@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"04721ebf6b408f2c7ce9cfe45e8e0cde2fef1777","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-25 15:26:43.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2cf8a4de46e94160a1b80b80b5c6ac6b : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/3c1fe4fd3561442da744c93199174c11 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5bcd75c773f7e89fc6dcbc6a5e16d654ae762043","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-25 15:44:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/25/805825/2/check/iscsi-driver-primera/e7a5581/e7a5581.html : FAILURE in 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1899affc3ea9bc0b0ad867af03c6901f0ee56aeb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-25 16:12:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/25/805825/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5d11001/5d11001.html : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":2},{"id":"17af3fcf4ef2e8c06119e1fe4c23f125a86a319a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-25 16:14:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/805825/2/check/iscsi-driver-nimble/f06871f/f06871f.html : FAILURE in 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0cb88eb15b5a457d2703f15b54aaf29702c54a64","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-25 16:46:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/805825/2/iscsi/ : SUCCESS in 1h 35m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/805825/2/smb/ : SUCCESS in 1h 08m 45s","accounts_in_message":[],"_revision_number":2},{"id":"49d7987c4337db8a2dae2b681a712b6e1a45a99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:34:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09ddee43bd34a6faf3f5b99f05da71d : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a33296f9aff47ef911869f7b733f369 : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc50ce2b67449d4b53089e407ed8aaa : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f41047233624c99a94b10a22ae34026 : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/c3d77e5ac1c44a12a9e4ca2c1ef03699 : SUCCESS in 1h 02m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3097ca65bb184ae187bdaf9a2931ebdd : SUCCESS in 1h 16m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d331cc8bbe364370b58d41a5b91e9185 : SUCCESS in 14m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1dad17ffc5b411da3abe3d772258041 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/729fceb95ee74fe795ea7293acc0175a : SUCCESS in 6m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a08527ef010a4b8c86d405cf6c9e4d24 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/77bea49d39484751a0bf4b6ef05b6604 : SUCCESS in 53m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/997491ae9b3541a29fbaf2964a07cef0 : FAILURE in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1db405a2d4d84999850e4cbf946bd570 : SUCCESS in 1h 19m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6095e63b818845999b50e1d50f04c82c : FAILURE in 1h 43m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8912cf4f8284610a7ffa98fff0ba57a : SUCCESS in 1h 26m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c10dfa8aac243ccb94c9fca0baaa44a : SUCCESS in 1h 09m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5cabf8f82c384ec18b2c37c465cbeff7 : SUCCESS in 1h 24m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2e3512f558e24a1ab3b7b6813c35758a : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c0ceeab118634cf5ad0749076ecb27b5 : SUCCESS in 2h 21m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a9d910b87424d7ea8b5f9df5a0dda3a : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3938b4292ffef0e610e275cb61ca4c0e1234ff84","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-25 17:41:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/25/805825/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aef9675/ : SUCCESS in 2h 30m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/25/805825/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6b8d2e2/ : SUCCESS in 1h 40m 28s","accounts_in_message":[],"_revision_number":2},{"id":"a61377eb502a6ae9c82cae9f1a7ae0f4525c415d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-25 18:17:07.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-2/ns5_nfs : FAILURE in 1h 15m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-2/ns5_iscsi : FAILURE in 1h 15m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f4eb4aced0ec92cfe8509c4dc90b7a4f57ad6d4b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-26 12:48:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/25/805825/2/check/dsvm-tempest-cisco-zonemanager-job/1984898/ : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2dc4f3211339f14a322d854434ecd09e1a973304","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-09-02 15:00:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/805825/2/20210902205602/NEC-ISCSI : SUCCESS in 2h 40m 28s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/805825/2/20210902205602/NEC-FC : SUCCESS in 3h 04m 23s","accounts_in_message":[],"_revision_number":2},{"id":"29e609fa650d2c955d4c3531ae45b859d3e34712","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-06 20:05:42.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-2/ns5_nfs : FAILURE in 39m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-2/ns5_iscsi : FAILURE in 38m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"945704ff24559be920b805f23cb38ee32ac30c2c","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-09-10 12:26:58.000000000","message":"Patch Set 2: Code-Review+2\n\nThese are good to get backported!","accounts_in_message":[],"_revision_number":2},{"id":"986d276bb2dc190fda0a95bf402d9c514c6da13d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-15 07:43:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMerged in wallaby (which is a part of cinder 18.1.0 release), looks appropriate for victoria as well.","accounts_in_message":[],"_revision_number":2},{"id":"eaf3bd9d3dadcf4326b5c53fdd5d63331a14cafe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-09-15 18:11:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"354a9b6e978b26906071c8de2672b5a9ae1168ac","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-09-15 18:38:54.000000000","message":"Patch Set 3: Workflow+1\n\nShould be good to go.","accounts_in_message":[],"_revision_number":3},{"id":"6f06265307ef0c5614cecb68a5b31c199e96299d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-09-15 18:52:20.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/4b251b95c3a545e883b5acefd1070db7 : RETRY_LIMIT in 29s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f1ba7e33aaee448a9624a9f38b8c23c8 : RETRY_LIMIT in 25s","accounts_in_message":[],"_revision_number":3},{"id":"74f61f2d98d69ed6327d8f6c74efdc4c64bfa3b1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-15 19:39:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/805825/3/iscsi/ : SUCCESS in 1h 05m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/805825/3/smb/ : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b300016d5134844513b67a21333ffbacc7f7b082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 20:22:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef30426325654649894855557c8f12bf : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6897f1177f574b8f80d58eb403109da8 : SUCCESS in 9m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b375af566ce7401082ca416b2ffa98ae : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954c02608c2e480d8d39d2cc9b288d49 : SUCCESS in 15m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/d497bed428c0413eba2846afb69e8b5e : SUCCESS in 1h 09m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/056ea8b4194e40e9890863ee3d74bf0b : SUCCESS in 1h 28m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a9d6cf8ce78e4cc2870545fcf6c1a0bb : SUCCESS in 13m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3da09de11eca4ae181f3a8f406eedac8 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b69cbfcc6a0647cd85661e765064e5ec : SUCCESS in 6m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3741d216c3b541918c2ffd9df552969d : SUCCESS in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/63329a5187f0413d837837d23df0cdb9 : SUCCESS in 43m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d59d3bf7b39e40e4a8710c41aa6bce30 : FAILURE in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/39f963c3b3c149fe82841173e37d7176 : SUCCESS in 1h 16m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa035760a10c47c1a90266ac9eb828d8 : FAILURE in 1h 41m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0dce52e29e384a24bc42ddfa66db4c99 : SUCCESS in 1h 41m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/614f30837f7a4216815e1abbe452435e : SUCCESS in 1h 00m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cf12b3bd411546b2a3e9a80267316b0a : SUCCESS in 1h 43m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/14834455dbec4df2a7255a25e9a06326 : SUCCESS in 1h 21m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60f4a6813a5c42fca7c9f3ed3fc1e422 : SUCCESS in 2h 09m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/551234f5a1fe4a30b072839f174c1cce : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":3},{"id":"73af9704c41679195e4548fe7989463b12352557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 20:40:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"75783ec21faf02948248770135db965d51c65ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 22:03:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb91803b84f42a18120cbd7753760c9 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88aa7b9618484ddd9d78f256262b9848 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d79164d84d9447c68466e68ccf380720 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62055e068813450b8d757d06e613df61 : SUCCESS in 15m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/42d97be68cb74d37968c032580e0092b : SUCCESS in 1h 05m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48708b685bc54e9bbc30e5ad34e93314 : SUCCESS in 1h 07m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80573113b98c4490913a5e042d0b3f76 : SUCCESS in 1h 08m 35s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/480265a2bb224d4c856612a4c4bab754 : SUCCESS in 1h 14m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93ec1bbc381e44aca96f233c2390e3f7 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":3},{"id":"11dcf088a1e27fe8965d7dfb330f2275ed2fffc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 22:03:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1235fb6bb131bf60e0f00007eb2274aeb8a23f93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 22:06:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0d72971d564818874e63b50f8940ac : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d3b1ca03967461fcd09711434e9762f0ac1d9d0a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-16 02:53:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k iscsi-driver-alletra-6k : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"4382ff234c0eefb0ff1ea2252299cdf2b00a0c80","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-16 02:55:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/25/805825/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a62df66/ : SUCCESS in 2h 33m 13s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/91d72a1143354521b225f4efce73e6f6 : DISK_FULL in 2h 53m 13s","accounts_in_message":[],"_revision_number":3},{"id":"ca2dff5529f7692537dcf0c363a5b45c7dffcfbc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-16 05:23:34.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-3/ns5_nfs : FAILURE in 36m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-25-805825-3/ns5_iscsi : FAILURE in 36m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"6a8e55da8a082cbab9ff3e786cfd85f0083b23cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-16 21:04:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/25/805825/3/check/dsvm-tempest-cisco-zonemanager-job/32bc396/ : FAILURE in 1h 20m 48s","accounts_in_message":[],"_revision_number":3},{"id":"ec7a3a43515a5f003efc9985bbacdf039366bc48","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-17 00:07:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/805825/3/check/iscsi-driver-nimble/e4e081a/e4e081a.html : FAILURE in 1h 12m 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7c34584af53b01bd359e1df37e3704288ebe30e","revisions":{"27d6bd331755c4ab72dacf6be9b6d205bc560ba2":{"kind":"REWORK","_number":1,"created":"2021-08-24 17:13:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/25/805825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/805825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/805825/1"}}},"commit":{"parents":[{"commit":"7c71e148893ce73d7267d78928631fe6147e0d35","subject":"Merge \"[SVF]:Storwize hyperswap volume is not deleting.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c71e148893ce73d7267d78928631fe6147e0d35"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-18 13:19:58.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-24 17:13:51.000000000","tz":0},"subject":"LVM: Retry lvextend commands on code 139","message":"LVM: Retry lvextend commands on code 139\n\nRetry lvextend commands upon segfault, similar to other\nLVM calls.  This affects the volume extend path.\n\nChange-Id: I0c0cb5308246a3dce736eade67b40be063aa78bb\nRelated-Bug: #1901783\nRelated-Bug: #1932188\nCloses-Bug: #1940436\n(cherry picked from commit c4b8956763c2759bd2b9fe511efac5b3edfb226d)\n(cherry picked from commit 2425f3ef590524c357622748df79d87edebd17f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27d6bd331755c4ab72dacf6be9b6d205bc560ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27d6bd331755c4ab72dacf6be9b6d205bc560ba2"}]},"branch":"refs/heads/stable/victoria"},"6dcccc50a5df7893811fce81c5873f394b39f4ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-25 14:56:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/25/805825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/805825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/805825/2"}}},"commit":{"parents":[{"commit":"2923c2ed10fadb677f2fa8393298a7f918ce2d45","subject":"LVM: Retry lvdisplay and lvcreate calls on segfault","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2923c2ed10fadb677f2fa8393298a7f918ce2d45"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-18 13:19:58.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-25 14:51:06.000000000","tz":-240},"subject":"LVM: Retry lvextend commands on code 139","message":"LVM: Retry lvextend commands on code 139\n\nRetry lvextend commands upon segfault, similar to other\nLVM calls.  This affects the volume extend path.\n\nChange-Id: I0c0cb5308246a3dce736eade67b40be063aa78bb\nRelated-Bug: #1901783\nRelated-Bug: #1932188\nCloses-Bug: #1940436\n(cherry picked from commit c4b8956763c2759bd2b9fe511efac5b3edfb226d)\n(cherry picked from commit 2425f3ef590524c357622748df79d87edebd17f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6dcccc50a5df7893811fce81c5873f394b39f4ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6dcccc50a5df7893811fce81c5873f394b39f4ea"}]},"branch":"refs/heads/stable/victoria"},"a7c34584af53b01bd359e1df37e3704288ebe30e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-15 18:11:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/25/805825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/805825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/805825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/805825/3"}}},"commit":{"parents":[{"commit":"ee1674b5a806998514db83cb771c1781b0966716","subject":"LVM: Retry lvdisplay and lvcreate calls on segfault","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee1674b5a806998514db83cb771c1781b0966716"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-08-18 13:19:58.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-09-15 18:11:25.000000000","tz":0},"subject":"LVM: Retry lvextend commands on code 139","message":"LVM: Retry lvextend commands on code 139\n\nRetry lvextend commands upon segfault, similar to other\nLVM calls.  This affects the volume extend path.\n\nChange-Id: I0c0cb5308246a3dce736eade67b40be063aa78bb\nRelated-Bug: #1901783\nRelated-Bug: #1932188\nCloses-Bug: #1940436\n(cherry picked from commit c4b8956763c2759bd2b9fe511efac5b3edfb226d)\n(cherry picked from commit 2425f3ef590524c357622748df79d87edebd17f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7c34584af53b01bd359e1df37e3704288ebe30e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7c34584af53b01bd359e1df37e3704288ebe30e"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
