)]}'
{"id":"openstack%2Fcinder~806011","triplet_id":"openstack%2Fcinder~master~Idb71b9a530b33a83c4e5efd2023533719ee65192","project":"openstack/cinder","branch":"master","topic":"1941068","hashtags":[],"change_id":"Idb71b9a530b33a83c4e5efd2023533719ee65192","subject":"Change the host config option type","status":"ABANDONED","created":"2021-08-25 15:19:30.000000000","updated":"2021-08-26 06:45:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e23248c1b5fea36792a4680e9467dcbf97778bf4","_number":806011,"virtual_id_number":806011,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 21:04:31.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-26 00:13:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"bd96ecdb96dee5683a6bc15851d6bea75454da9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-08-25 15:19:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c417c0e88112393cb87cda1f20cc1f74072acd5f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-08-25 15:22:01.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/cinder/+/806010 which came in first.","accounts_in_message":[],"_revision_number":1},{"id":"23339e0c8f39038068a5363ec54da62c84bcb3b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-25 21:04:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/11/806011/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1e6c7c958c5767e7efb557feead96762d3ee99b2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-26 00:13:31.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-11-806011-1/ns5_nfs : SUCCESS in 4h 6m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-11-806011-1/ns5_iscsi : SUCCESS in 4h 12m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e23248c1b5fea36792a4680e9467dcbf97778bf4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-26 06:45:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/11/806011/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e774e8152d09c4209d971028ae73ddd545befced","revisions":{"e774e8152d09c4209d971028ae73ddd545befced":{"kind":"REWORK","_number":1,"created":"2021-08-25 15:19:30.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/11/806011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/806011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/806011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/806011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/806011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/806011/1"}}},"commit":{"parents":[{"commit":"a2f0f57a06e80aa724233d60ec239b7d44e2746f","subject":"Merge \"db: Reorder initial migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2f0f57a06e80aa724233d60ec239b7d44e2746f"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-08-25 15:17:07.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-08-25 15:17:07.000000000","tz":120},"subject":"Change the host config option type","message":"Change the host config option type\n\nThe `host` config option is currently defined as `HostAddressOpt`, yet\non the other hand the `backend_host` option is defined as a `StrOpt`.\n\nAccording to the `host` option this is an opaque identifier, so this\npatch changes the option to a string.\n\nThis fixes issues with `cinder-manage db sync` command.\n\nCloses-Bug: #1941068\nChange-Id: Idb71b9a530b33a83c4e5efd2023533719ee65192\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e774e8152d09c4209d971028ae73ddd545befced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e774e8152d09c4209d971028ae73ddd545befced"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
