)]}'
{"id":"openstack%2Fcinder~806605","triplet_id":"openstack%2Fcinder~master~Ic61f23c87de5dad52c06061849e3db970cee4327","project":"openstack/cinder","branch":"master","topic":"bug/1942095","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-08-26 09:16:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"34290":{"account":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"last_update":"2022-08-05 09:37:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34290\u003e replied on the change","reason_account":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2022-09-15 19:25:01.000000000","reason":"Change was submitted"},"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2022-09-15 19:25:01.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-08-26 06:04:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-08-05 09:10:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9535\u003e replied on the change","reason_account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}}},"hashtags":[],"change_id":"Ic61f23c87de5dad52c06061849e3db970cee4327","subject":"Fix volume caching in PowerFlex driver","status":"MERGED","created":"2021-08-30 14:56:56.000000000","updated":"2022-09-16 18:15:14.000000000","submitted":"2022-09-15 19:25:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806605-bug/1942095","meta_rev_id":"0ffa57b21fff7b430ad26ec8951badac54b1fef5","_number":806605,"virtual_id_number":806605,"owner":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-15 19:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2022-08-26 09:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2022-09-15 18:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":1,"date":"2022-08-26 12:45:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":2,"date":"2022-09-05 10:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2022-09-15 18:13:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":34071,"name":"Alexey Sedlyarsky","display_name":"Alexey Sedlyarsky","email":"palyla@ya.ru","username":"palyla"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":34071,"name":"Alexey Sedlyarsky","display_name":"Alexey Sedlyarsky","email":"palyla@ya.ru","username":"palyla"},{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 14:58:11.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2021-08-30 14:58:11.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-08-30 15:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 16:07:55.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-30 16:09:40.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-08-30 16:09:40.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-30 16:41:57.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-08-30 17:38:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-30 18:09:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 18:15:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 18:39:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 18:50:38.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 18:52:01.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 19:27:53.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-08-30 19:47:42.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-08-30 19:57:06.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-08-30 20:19:41.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-08-30 21:24:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-30 23:02:19.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-08-30 23:40:56.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 23:43:17.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 23:56:33.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 04:35:08.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-08-31 07:03:37.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2021-09-01 23:12:37.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-02 02:25:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-07 04:25:49.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-14 10:45:35.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-09-30 12:43:47.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":34071,"name":"Alexey Sedlyarsky","display_name":"Alexey Sedlyarsky","email":"palyla@ya.ru","username":"palyla"},"state":"REVIEWER"},{"updated":"2021-10-01 11:32:52.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-01 13:47:57.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-10-03 06:53:27.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-03-03 09:38:27.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-03-03 09:39:00.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-03 11:00:57.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 09:17:21.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2022-06-15 09:17:21.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-06-17 09:48:36.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-06-17 10:54:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-17 15:48:30.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-21 11:31:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-06-21 15:53:21.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2022-06-29 08:16:31.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"state":"REVIEWER"},{"updated":"2022-07-06 15:41:07.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-07-06 16:01:28.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-06 19:33:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-12 12:06:09.000000000","updated_by":{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"},"reviewer":{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"},"state":"CC"},{"updated":"2022-07-18 20:49:22.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-11 14:00:33.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-08-26 06:04:51.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-08-26 08:24:27.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-26 14:17:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-15 14:08:20.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"}],"messages":[{"id":"69327e79323bb72c4d4a33fc24231ce49e3b0113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-08-30 14:56:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51f75c2259aab9c4d40af0a287e19650b437e399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-08-30 15:25:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f92bc805d82171c52c94bbdbfb4c93e0cc23180c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 15:50:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8eeebe734414b4ea5b3ee8b117723ed : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f145ea695955428e9cc4123a1ac7e91e : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c66422d021988f4723f0ab5c17b13b09c1a7405","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-30 16:41:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/2/iscsi/ : SUCCESS in 58m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/2/smb/ : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2d17b8b4bb5f3fa8da3c942d9cad659e1e321a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:38:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43dbd7dda4fb42d2ba932444498fa834 : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62019e0c868440eba83763a08bedeafc : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ab6798856fa482cabf250793b758e0d : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a2af7f82de74c77ba8ac0df90a7cfa3 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d993c6245a445288f4880f9cb7cf74 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/9a6cb01b29344c2ca21f2d71f15b05cc : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d2a0d510c0041c3afb6babe363f0a7b : SUCCESS in 1h 09m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e3c68a4c5744b69874ee1e08258ddf7 : SUCCESS in 12m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2ba95f4a00545b6b65a91e86d313c71 : SUCCESS in 7m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51463aaa86734c939c00931590087cc0 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/091a58204729477690c0bab5d0cd7a56 : SUCCESS in 7m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c201b52ba46b4395aaabf0070790d98c : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/70e21e0bbb07484eb132791395298a68 : SUCCESS in 44m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/724603328e6f474289cdde10eba94a70 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2a289ea8cd34e46b36825d8e705f62e : SUCCESS in 1h 20m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00484e4055544a17b67274f9a11be5b0 : SUCCESS in 1h 40m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f24bcfd473864b579cdc08510f648472 : SUCCESS in 1h 27m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b775867949e34cc7a3b12e7c90d1779c : SUCCESS in 1h 08m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90fba09eca234f9fbdbb2d5fee0760a3 : SUCCESS in 1h 11m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/40c5647253c7492aa67a9a36498d70dc : FAILURE in 1h 17m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94d195025fb44a85b6f851f9a3d84e2c : SUCCESS in 2h 08m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82bbbd25238a4e489afa7818a83e79f5 : SUCCESS in 45m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb8a7e15eb664520859fccf4781b5e7f : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":2},{"id":"dc3f9b4cb68297d35e28631fa7f1d5a79abef56c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-30 18:09:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/806605/2/check/cinder-storpool-tempest/2787ca4/ : RETRY_LIMIT in 1m 52s","accounts_in_message":[],"_revision_number":2},{"id":"28e9f3a0f14753aeb2a05b5370c84e789cd7e65e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-30 18:15:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/806605/2/check/iscsi-driver-primera/95d052b/95d052b.html : FAILURE in 34m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a088886357a1bc44745c10e08c146aca32e85c4f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-30 18:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/05/806605/2/check/iscsi-driver-alletra9000/9f647a0/9f647a0.html : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a6f3b4b9877619ee8484f0a76688f7a67366c55d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-30 18:39:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/806605/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7610cfebb587a4b210d1c07fd8d7e94a95e1baf6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-30 18:50:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/806605/2/upstream-check/cinder-cDOT-iSCSI/7144f3d/ : FAILURE in 56m 08s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/806605/2/upstream-check/cinder-cDOT-NFS/3a66f2e/ : SUCCESS in 36m 45s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/806605/2/upstream-check/cinder-cDOT-FCP/ac70972/ : FAILURE in 1h 10m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"03c9b4245b97b847d529e941f0e43b11c896bac5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-30 18:52:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/2/check/DellEMC_VNX_iSCSI/3022775/DellEMC_VNX_iSCSI/365 : FAILURE in 1h 13m 10s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/05/806605/2/check/DellEMC_VNX_FC/f680189/DellEMC_VNX_FC/388 : FAILURE in 53m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"4a83241dd533747a774850180c229fe8019b085d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-30 19:00:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/806605/2/check/iscsi-driver-nimble/783c5d5/783c5d5.html : FAILURE in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ed1fd414f9cc8ebc88cac6cf815f1228c623cf06","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-30 19:27:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/806605/2/upstream-solidfire-check/cinder-solidfire-iSCSI/4f0f16d/ : FAILURE in 1h 47m 24s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2a86b6a53f1c0b0c88ed89194d3fb69096e0543","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-08-30 19:47:42.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/806605-2/HPMSA-iSCSI : SUCCESS in 1:04:42\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/806605-2/HPMSA-FC : SUCCESS in 32:53.00","accounts_in_message":[],"_revision_number":2},{"id":"aa1bef6903fc4db8b835a4da1c8b2875bcaf2761","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-08-30 19:57:06.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/806605-2/Seagate-iSCSI : SUCCESS in 1:09:40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/806605-2/Seagate-FC : FAILURE in 37:50.58","accounts_in_message":[],"_revision_number":2},{"id":"bccec84bea66fd7f47fc34a96fb781ea73331dd3","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-08-30 20:19:41.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/806605-2/PVME-iSCSI : FAILURE in 9:21.06\n* PVME-FC http://os-logs.tristero.net/dell-ci/806605-2/PVME-FC : SUCCESS in 43:08.91","accounts_in_message":[],"_revision_number":2},{"id":"451a6902150d2a17750f2febff8848f40e642bbe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-30 20:39:26.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/806605/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"50f1da5c03041e89f2d0f504d6b9628e203dabc1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-30 21:24:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/806605/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e2c9b57/ : SUCCESS in 2h 36m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/806605/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cbf0ddc/ : SUCCESS in 1h 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c19227697c12f3ef5bdd8ac87753d65c02d2754c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-30 23:02:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/806605/2/check/dsvm-tempest-cisco-zonemanager-job/b0f94d7/ : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8d78193d3ff9cbe46d239f3a35acf51f449ea529","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-30 23:40:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/05/806605/2/check/DellEMC_XtremIO_FC/7e74359/DellEMC_XtremIO_FC/825 : SUCCESS in 1h 16m 39s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/2/check/DellEMC_XtremIO_iSCSI/941c655/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"9f2d275bae096b2c085de5a10cb55b7397d3b24a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-08-30 23:43:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/806605/2/ : SUCCESS in 1h 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e8c4fb6c9553c428ff9a391c9d234288ce920dc4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-30 23:56:33.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-1/ns5_nfs : FAILURE in 53m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-1/ns5_iscsi : SUCCESS in 4h 10m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d5a25f5b25c9652ffefa3dff51edd70e5286e180","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-08-31 04:35:08.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/806605-2/Lenovo-iSCSI : SUCCESS in 1:20:16\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/806605-2/Lenovo-FC : SUCCESS in 52:40.46","accounts_in_message":[],"_revision_number":2},{"id":"302399003b68a3f5cd9adc380486827eb064ec57","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-31 04:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-2/ns5_nfs : SUCCESS in 4h 17m 13s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-2/ns5_iscsi : SUCCESS in 4h 23m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"b931dfa718793922bcfca59f8bb131cb549f556a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-31 04:39:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/806605/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"88a13f2a08195a9180e0b91ecb43998790731f29","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-08-31 08:36:58.000000000","message":"Patch Set 2: Code-Review-1\n\nCould a unit test be added please.  And possibly a release note, but I will defer to the another reviewer to confirm.","accounts_in_message":[],"_revision_number":2},{"id":"33f12c987ecbd03dffb2efd491aafadc60a644db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-31 09:15:10.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/806605/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2d89fac53f55a06ba6fbb9e2d52edec904bb655d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-08-31 12:27:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1568bd71e5151b334635593cf5fbbace5f7bf974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:51:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20ad9e3663fe4dfa86eeb41530e821d3 : SUCCESS in 21m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbdbf708cd2c4161b098acea490f6de4 : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"11c161114ba59b2475ac04b8aa355d0e4eedcbf4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-08-31 13:52:56.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/3/iscsi/ : FAILURE in 57m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/3/smb/ : FAILURE in 59m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e57ef3feba66105a99b64e3156021a45575e791e","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-08-31 13:58:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d127f4bc2b6ad70545d3addd4c87d14e6b34b021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:42:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6422c0129c754f85ae5647592a8f62ee : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39a9b3abf7c945abade462186e39fd29 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46683f9def284b6c9f638caca96d727d : SUCCESS in 10m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ec53456719f4f94a9089f2b69f527d5 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ada84d7f09b480d83c3023ca9076ef9 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/bc388636ef8c41658c65b11fe3ab388b : SUCCESS in 1h 18m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/611d672ccfcf40e58998485b05d9903d : SUCCESS in 1h 08m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f70f61e3812045e2ba94fa5be371eca0 : SUCCESS in 7m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7d6ed927faa4428970a0e5f777050ce : SUCCESS in 12m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab0e6d6d72394c28a98ea73e2cfdeeaa : SUCCESS in 6m 32s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6176b62c1ec472ab1eb772b793f0548 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae42e097160b46b5bdb73de2e7c5fe99 : SUCCESS in 7m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ce80f10c5ad4136a720742fc3547ac2 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a5a9e15f55a47a78f2a2b2c55ff6791 : FAILURE in 42m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/260e28ed1a3c41f3a61aa54f8e631d2e : SUCCESS in 8m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57006ec0fade4e5fbd256cd915a62227 : SUCCESS in 1h 26m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3a8c1c0b0c8a4029a35ce1c1f519d53b : SUCCESS in 1h 36m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f9a382bc46c74c4f9ff216164ac77926 : SUCCESS in 1h 29m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6730d37e731a46948347baa3971e8091 : SUCCESS in 1h 15m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/24b8032362b54d48910a8888ae5717d9 : SUCCESS in 1h 19m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6117ff2fb6ab4673ab0cf77c446d68c0 : SUCCESS in 1h 11m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bfe17fa5e50d4c519cfbbf84b93503a9 : SUCCESS in 2h 09m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b4d9e7615f04b2780d529adb163faf7 : SUCCESS in 53m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0612bd9b0f9c43a2a218b60ce2ffaa55 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":3},{"id":"12779ef1d3f9cf1c9a4ff636c17c883099e145f9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-31 15:53:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/806605/3/check/iscsi-driver-primera/b0f6b82/b0f6b82.html : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":3},{"id":"afb2d999d9aeb201313cff9684f4d9633b7beee2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-31 16:25:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/806605/3/upstream-solidfire-check/cinder-solidfire-iSCSI/27d83e9/ : FAILURE in 1h 40m 59s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"67bfccbefe51a0685d95eaa66408d957d4f5b11a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-08-31 17:24:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/05/806605/3/check/iscsi-driver-alletra9000/6b72329/6b72329.html : SUCCESS in 1h 21m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d616c4a5d67f754b1025c5ac59336ed562bdb9b1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-31 17:54:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/806605/3/upstream-check/cinder-cDOT-iSCSI/1658aa7/ : SUCCESS in 33m 37s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/806605/3/upstream-check/cinder-cDOT-NFS/ec5a49e/ : SUCCESS in 34m 27s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/806605/3/upstream-check/cinder-cDOT-FCP/dc6ea4c/ : FAILURE in 2h 11m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2e31eefc75bf3e179883e5ad7bab74957162fd28","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-08-31 18:02:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a1dc28b5ec515af04faca7a25deb9253063827b9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-01 00:24:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/806605/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0eeb8d9/ : SUCCESS in 2h 14m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/806605/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a0248ca/ : SUCCESS in 2h 07m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b4e00cf48993669e06031ef71addac52a7278c4f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-01 02:47:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_VNX_iSCSI/0402eae/DellEMC_VNX_iSCSI/389 : FAILURE in 1h 12m 51s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_VNX_FC/a47c4a5/DellEMC_VNX_FC/414 : FAILURE in 48m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"f3539a5973d7290e57ca5b5261aed75717f03436","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-01 04:32:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/806605/3/check/iscsi-driver-nimble/eb3a380/eb3a380.html : FAILURE in 1h 33m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b7e5546c18bcc3199502876a51496bda2d32e065","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-01 09:42:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/806605/3/check/dsvm-tempest-cisco-zonemanager-job/1218ae8/ : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":3},{"id":"87004744c067df4ec19ae1274bee8a5a20925103","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-01 22:21:33.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-3/ns5_nfs : FAILURE in 18m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-3/ns5_iscsi : FAILURE in 18m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"79440c5262148d8d816775042dd9af9845b2fd09","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-01 23:12:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_PowerFlex/0109186/DellEMC_PowerFlex/482 : FAILURE in 1h 36m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"b10be6348ff0f3c9a2b8308f84ead9c46cb82871","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-09-02 00:22:43.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/806605-3/HPMSA-iSCSI : SUCCESS in 1:02:16\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/806605-3/HPMSA-FC : SUCCESS in 33:32.69","accounts_in_message":[],"_revision_number":3},{"id":"bb44a5c0fa247e68c21968be2a6974e98d6dc8b3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-02 02:25:13.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/806605/806605/3/check-not-vote/ext-cinder-zuul/62ead1f : FAILURE in 33m 29s","accounts_in_message":[],"_revision_number":3},{"id":"5eb9fe4e81cd3a7121bbf2ebd4c39b9c2660f79d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-09-02 05:55:27.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/806605-3/Seagate-iSCSI : SUCCESS in 1:11:20\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/806605-3/Seagate-FC : SUCCESS in 40:19.83","accounts_in_message":[],"_revision_number":3},{"id":"392a3d06930aafa01641aa5135146cc590f52f06","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-09-04 03:40:22.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/806605-3/Lenovo-iSCSI : SUCCESS in 4:15:20\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/806605-3/Lenovo-FC : FAILURE in 2:51:12","accounts_in_message":[],"_revision_number":3},{"id":"99a27190b9dce374983f52e2a68ade1a820d3453","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-09-07 04:25:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/3/20210907101452/NEC-FC : SUCCESS in 3h 06m 56s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/3/20210907101452/NEC-ISCSI : SUCCESS in 3h 10m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3aacb0ce554e9ff4d81d43207b7f33d9456020c4","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-09-07 11:11:30.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerFlex CI","accounts_in_message":[],"_revision_number":3},{"id":"458cd50c12f6cb44b70346917b5364e2ca902e4c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-07 11:16:35.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-3/ns5_nfs : FAILURE in 4h 25m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-3/ns5_iscsi : SUCCESS in 4h 27m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"af38a2f76f9942169189ff4f965470bda86e860b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-07 19:08:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_PowerFlex/417127b/DellEMC_PowerFlex/527 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"0c21338b34b5f21afe7db878217bfb407d62d0c1","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-09-14 10:45:35.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"d082f075010887dad2f88c0019e750736903f7e2","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-09-14 10:46:40.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"b61220ad9efcf5f654997ab35b3f20d9ce3d9b91","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-09-22 06:56:48.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerFlex CI","accounts_in_message":[],"_revision_number":3},{"id":"8ea3d93c4af1e9dd167784bf5d9c0cf9d588b5ba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-22 08:49:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_PowerFlex/1c23ef9/DellEMC_PowerFlex/681 : FAILURE in 1h 51m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"80e0c34db5e468a6ba9cddea90e12ce2ff578517","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-09-22 09:54:49.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerFlex CI","accounts_in_message":[],"_revision_number":3},{"id":"a381104d1b2d1eeb805766b4f279d48811d5e24c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-22 13:13:24.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_PowerFlex/a3bfccc/DellEMC_PowerFlex/683 : FAILURE in 1h 48m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"3a6a24c925d3894d70b53db3e878c6633914cf3c","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-09-23 03:38:21.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerFlex CI","accounts_in_message":[],"_revision_number":3},{"id":"89858d50dc32b896936a4080697fac766b827f44","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-09-24 09:46:51.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerFlex CI","accounts_in_message":[],"_revision_number":3},{"id":"659fba08d193217847f2f18e80ed1472dbb290de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-24 12:35:39.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/3/check/DellEMC_PowerFlex/13f2e0f/DellEMC_PowerFlex/695 : SUCCESS in 1h 30m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"86b77e05d6305bc393dfcf69a81d81f9d39da3f3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-25 11:23:29.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-3/ns5_nfs : SUCCESS in 4h 27m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-3/ns5_iscsi : SUCCESS in 4h 34m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8ed9a093a1b859e69245750ddaf963094c16f5eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-10-01 10:07:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1a36eb1d4995dac7019dfd7533352880846df377","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-01 10:07:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"fd412c83fb41fe79cbeb756352e26eaf1392e020","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 11:12:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5768046098834e3abc950334de6c4f31 : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46ae402065344624bbc4080942c64ec2 : SUCCESS in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"acf5b4be2d6db89739e9c7a08c1abe9f920c62c0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-01 11:17:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_XtremIO_FC/bc5ca0b/DellEMC_XtremIO_FC/1079 : FAILURE in 1h 09m 40s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_XtremIO_iSCSI/d4e878d/DellEMC_XtremIO_iSCSI/954 : FAILURE in 30m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"04f8bb6f231f2b305547792ff654f6accc71f348","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-01 11:32:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_Unity_FC/341362b/DellEMC_Unity_FC/994 : FAILURE in 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_Unity_iSCSI/e799c0d/DellEMC_Unity_iSCSI/859 : FAILURE in 1h 23m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"652221977e694e4a94de8659d1fbba3e08b264f6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-01 11:35:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/05/806605/4/check/iscsi-driver-primera/a4159db/a4159db.html : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":4},{"id":"116cfab4d466d9e3e9948c895c136ca82ffd4ce9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-01 11:41:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/806605/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9afd497/9afd497.html : SUCCESS in 1h 17m 46s","accounts_in_message":[],"_revision_number":4},{"id":"12eaec503eec9e5782e30d76cc043086ed37a69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 12:03:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6a334d47324b1ba88243a6530b1b8b : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eedaeaf32d074f0bb73dbbeb53b9403d : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30ee6af64afd4fe4ad92755f3143047c : SUCCESS in 13m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cb4277928384797ab8f9b8f98b234ff : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd6e7379ff34e2bba87d08a2eb1a4f8 : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3ca33af60c054c1f9a061cdf1bae50a7 : SUCCESS in 1h 00m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bd9646954864cefad5210c12053a145 : SUCCESS in 58m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebf2a2678a7e4e5097b82c804ab2f982 : SUCCESS in 7m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65d8e9c0913041f28a4e09aabab1a042 : SUCCESS in 17m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f21a0c415b247138e841991bf557b7e : SUCCESS in 5m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f86dd55c905b4693b68e43c2766fc8ce : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50af4a47f74c458d8c57c47219f260a4 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9913f007e144c4b9f1b1e4ba2304255 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/65b19309284347399b0d48abe066308e : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0202ad6cdcb94ce0ab40bd39bc96fb2d : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/291cc680776344b29348d68aba0fe341 : SUCCESS in 1h 08m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/885877205adf4011bd55adf6075a1675 : SUCCESS in 1h 28m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c97e9bd3c8904b7ba566aeed6401fe5a : SUCCESS in 1h 35m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/958e56b44e674ea194c5cbc19fe5635e : SUCCESS in 1h 01m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a39e664b262d4e7dbced07a28a780cc2 : SUCCESS in 1h 09m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca58faeecac64650a71cf4e00be18703 : SUCCESS in 39m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f067248bbf6d460d8b55bccab23aedc2 : SUCCESS in 1h 11m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b8593d208994372a14779fd5f18afe9 : SUCCESS in 1h 52m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea22f111491b4e6181c8d24c54af9529 : SUCCESS in 41m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a30bedb38dd40378bcc3dd6788fa8bd : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":4},{"id":"a053fa13b9279a528b72ab9b0a786dfad2f4cc5d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-01 12:05:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/806605/4/upstream-check/cinder-cDOT-iSCSI/8fa63a8/ : FAILURE in 43s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/806605/4/upstream-check/cinder-cDOT-NFS/b286994/ : FAILURE in 43s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/806605/4/upstream-check/cinder-cDOT-FCP/88b5339/ : FAILURE in 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"4b1943667911d2caa47409b7ae5d54227b92c193","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-10-01 12:05:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/806605/4/upstream-solidfire-check/cinder-solidfire-iSCSI/30462ee/ : FAILURE in 38s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"2d29b8efd105b753305378798f6781e9687a2196","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-01 12:22:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/806605/4/check/iscsi-driver-alletra-6k/d86b14b/d86b14b.html : FAILURE in 1h 39m 57s","accounts_in_message":[],"_revision_number":4},{"id":"1cc2d0b258b2181d3a1b3fed9ead86ecd622d2ff","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-01 13:13:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/806605/4/check/iscsi-driver-nimble/cb2ac1b/cb2ac1b.html : FAILURE in 1h 12m 34s","accounts_in_message":[],"_revision_number":4},{"id":"42441a4a29f3946df3574777a4de9c76fde9c555","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-01 13:13:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/806605/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d55b235/ : FAILURE in 2h 16m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/806605/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c500131/ : FAILURE in 3h 04m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4deb1d83f576899740b1cc2909f2f9af659b25d5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-01 13:41:34.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_XtremIO_FC/02aeaa7/DellEMC_XtremIO_FC/1081 : FAILURE in 1h 05m 22s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_XtremIO_iSCSI/6b15738/DellEMC_XtremIO_iSCSI/956 : FAILURE in 1h 37m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"5a8e5f681adf106f0b55665a2d68346c30810d6e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-10-01 13:46:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerFlex/beae7f1/DellEMC_PowerFlex/713 : FAILURE in 1h 43m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f3e728223bc19d566dda273921505bf71dce3356","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-10-01 13:47:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_SC_FC/ea5e9b9/DellEMC_SC_FC/629 : FAILURE in 48m 54s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_SC_iSCSI/a29f1c6/DellEMC_SC_iSCSI/597 : FAILURE in 1h 44m 10s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"52d884d52a928e4619a6dfa9ccfa22002abcf797","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-10-01 14:00:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/806605-4/PVME-iSCSI : SUCCESS in 44:34.10\n- PVME-FC http://os-logs.tristero.net/dell-ci/806605-4/PVME-FC : SUCCESS in 43:50.89","accounts_in_message":[],"_revision_number":4},{"id":"c27ebff12887b76893f244b1fc9b1cd707e66e74","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-01 14:04:12.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-4/ns5_nfs : SUCCESS in 3h 56m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-806605-4/ns5_iscsi : SUCCESS in 3h 54m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"21bb4ed0cf6ee6d8b1a07c5036659eaedf3744a8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-10-01 14:10:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/806605-4/HPMSA-iSCSI : SUCCESS in 55:48.45\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/806605-4/HPMSA-FC : SUCCESS in 39:21.67","accounts_in_message":[],"_revision_number":4},{"id":"bfb761e009d2cdb04f06654be17cf3915cba1edb","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-10-01 14:23:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/806605-4/Seagate-iSCSI : SUCCESS in 1:09:24\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/806605-4/Seagate-FC : SUCCESS in 39:27.85","accounts_in_message":[],"_revision_number":4},{"id":"e16792509f69e0a81e5e1c5c5186def0d3265a15","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-01 14:27:07.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_Unity_FC/2ac75e2/DellEMC_Unity_FC/996 : FAILURE in 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_Unity_iSCSI/7ee8b32/DellEMC_Unity_iSCSI/861 : FAILURE in 1h 24m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"5e771d95c8a63d39b06f3158cde638c7eb7ff040","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-10-01 14:33:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/806605-4/Lenovo-iSCSI : SUCCESS in 1:14:10\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/806605-4/Lenovo-FC : SUCCESS in 45:45.85","accounts_in_message":[],"_revision_number":4},{"id":"cd61c9d662f877dea1015bde4c6603ebe6a4aaa8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-01 15:43:09.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_VNX_iSCSI/3a86999/DellEMC_VNX_iSCSI/632 : FAILURE in 3h 34m 18s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_VNX_FC/60591ac/DellEMC_VNX_FC/621 : FAILURE in 1h 43m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"8c983d20a40c87389d294fb1759531a41dc02269","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-02 01:34:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/4/20211002074310/NEC-ISCSI : SUCCESS in 2h 48m 42s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/4/20211002074310/NEC-FC : SUCCESS in 2h 51m 39s","accounts_in_message":[],"_revision_number":4},{"id":"871e350a220486d8b5786214b4724aed598248ed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-03 06:50:30.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/4/iscsi/ : FAILURE in 44h 42m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/4/smb/ : FAILURE in 44h 41m 27s","accounts_in_message":[],"_revision_number":4},{"id":"ef8b29dcf132bc50cba114a52767e46171a84447","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-10-03 06:53:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerStore_iSCSI/afde2f1/DellEMC_PowerStore_iSCSI/None : RETRY_LIMIT\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerStore_FC/a870699/DellEMC_PowerStore_FC/736 : FAILURE in 1h 09m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"92cf09f221bec525bb0cab1797b2a0e29044680d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-10-08 14:50:43.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/806605/806605/4/check-not-vote/ext-cinder-zuul/c4d8ebb : FAILURE in 1m 49s","accounts_in_message":[],"_revision_number":4},{"id":"900166c2f473899f0bdf8edb592ef86b3ee8e101","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-03-03 08:26:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d0f8684129b277f6746bd7a20e427851245c876","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 08:51:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7428c418a70f4ffb92596617f6062201 : SUCCESS in 21m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/376e16fac565460f850ea8f15b8cc531 : SUCCESS in 22m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"505a62360b2f3a3eacd87f5d594e50faeb77c8a4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-03 08:52:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-solidfire-check/cinder-solidfire-iSCSI/7275a8f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"4a113753ae8ff176a93a26d69eb67da6371fd83e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-03 09:02:53.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/806605/806605/4/check-not-vote/ext-cinder-zuul/155be78 : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":4},{"id":"53e8dd898b2f67a4f0c17791960c281337d1c964","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-03 09:34:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/4/iscsi/ : SUCCESS in 59m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/4/smb/ : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":4},{"id":"1945e917fd519e5d6da597720e25de2ea05bd461","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-03-03 09:38:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/486ccf5/index.html : SUCCESS in 1h 09m 07s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/f507d3f/index.html : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":4},{"id":"90a447b758c2199b10a2522bcb6ae0ef4496acac","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-03 09:39:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/05/806605/4/check/dsvm-tempest-zadara-driver/3ff7b55 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a2d1ca9a46a0f88afb6e22ab7f44f3d4e6d20cd4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-03 10:31:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-cDOT-iSCSI/60d1596/ : SUCCESS in 1h 38m 46s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-cDOT-NFS/0e4a449/ : SUCCESS in 1h 34m 30s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-cDOT-FCP/a4762ed/ : SUCCESS in 53m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7f8dee4386922542e2c33822a22e34c20d1a9196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 10:48:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9646ada5e5049b290ebec5cc39bb32f : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb360b6554fc49799e6446f70d0f619b : SUCCESS in 12m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be1786a9a5554b6b8fb6af0c1a71912d : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bae1393d54472cbdc3e85577b2ee66 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/3e99997116a146a49b0c2ba7ee731477 : SUCCESS in 57m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c0a95b5cfde241c3b600b5399fad202a : SUCCESS in 49m 04s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a71522dfde894063a8f9ce02249569ae : SUCCESS in 1h 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfc73c830ba44bfd9c5b512cd9ef7e66 : SUCCESS in 52m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bdb776ab64749e9bbceb24c164ecb87 : SUCCESS in 7m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1d6f3b02aaee45dfb94149c406ff99fb : SUCCESS in 15m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ddb353ae21eb4531ac22df0fe2350d46 : SUCCESS in 6m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77b624648a1f471182221f0ea8fecf17 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e234b7aa7d6d4e7ea7887aba922267ba : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11cfaf56741442e6a1b06cea6319a2d0 : SUCCESS in 9m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad5ef7760a1b432dbb432f1018db2c44 : SUCCESS in 48m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/601f8d82fd634930bad36ce234a841d7 : FAILURE in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6f4cbe9a3748405097fbb307c7bfe6e4 : SUCCESS in 1h 26m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5ec969f4239541d6b90c57e881052163 : SUCCESS in 1h 15m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3fff3bac18094b8c98ac2276022bc482 : SUCCESS in 1h 35m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90dd032429fe4aa0b3529aec32eec0d6 : SUCCESS in 1h 07m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97a2707ea178444d9eab83b155de2348 : SUCCESS in 1h 19m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/50a2ce0670194e51affb746ee4a8cbd0 : SUCCESS in 45m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e09fe37e77bc4e2f882ada9c693c6de6 : SUCCESS in 1h 09m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08f238a8051043e39c02b7a54269c3d3 : SUCCESS in 2h 19m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72ac174dbd1a439b984530e871490ebc : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":4},{"id":"396febfd77a4972bb64c4edbd33a587b3820c8d3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-03 10:49:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-solidfire-check/cinder-solidfire-iSCSI/76d1f01/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"ac58d0593e2ccc85b306526aac9ec25e1d815428","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-03 11:00:57.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/806605/4/check/tempest-dsvm-full-focal-py3/215b266/ : FAILURE in 30m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/806605/4/check/devstack-plugin-ceph-dsvm-focal-py3/9ef9b4f/ : SUCCESS in 2h 27m 35s","accounts_in_message":[],"_revision_number":4},{"id":"77e58c9acb18ba60697a12f26ec277293c58784e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-03 11:53:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-cDOT-iSCSI/ef9bc0d/ : SUCCESS in 1h 03m 45s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-cDOT-NFS/1daa8b8/ : SUCCESS in 36m 36s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-cDOT-FCP/28f23d6/ : SUCCESS in 57m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a626c8dacc46363aace2787fa3304f8d7428014e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-03 14:52:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/806605/4/check/dsvm-tempest-cisco-zonemanager-job/3652f6a/ : POST_FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":4},{"id":"1ef303b7cbbd0be5ee5a464ea53f43b6917b1b85","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-06-15 09:18:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"66c27d9dcaf12133ad406f4388c3ef331f70542a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-15 09:21:01.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerFlex/5b00f98/DellEMC_PowerFlex/857 : FAILURE in 1m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"8bf6da75ca3d87403f6aa9b3f6cf0079c164e3f4","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-06-16 12:41:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67ab6c8fb043e89939606442571d2b7d00bfff86","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-06-17 08:53:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fc89d2b6a4a45c2d1adc1e3668e66f5ee26bf85a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 09:12:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a843ec0f0a54b8697dae55766dde137 : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78bab4bee5294531bfc4f92f93e7a722 : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e753457d7f74b6cae42880a60f21493e1c6dfa8b","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-17 09:48:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-05-806605-4/1655457161 : SUCCESS in 17m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"ed2112f943132f2f2caf6e70de830063138d9a1c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-17 10:32:18.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/4/iscsi/ : FAILURE in 1h 38m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/4/smb/ : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":4},{"id":"1ec524a85cb52c28a63fe9b7c8faa5807680f5c0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-17 10:54:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-tempest-plugin-ontap-fcp/69687bd/ : SUCCESS in 1h 39m 42s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/0e35bf3/ : SUCCESS in 1h 11m 10s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-tempest-plugin-ontap-nfs/9c8bba1/ : SUCCESS in 51m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"785b557ffef954699ff96ded54440c75c4aa4c39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-17 11:07:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/805420c996504cfb9911377b711adcf4 : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":4},{"id":"732d72c469b5616641cd1f48fd57115210d7341f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 11:07:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2665f686b04e7d8b836c3313b3baf4 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e85fb1f16143465db7347e937ae6dd1b : SUCCESS in 17m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/995df12f94ec4228be634531c2954dc2 : SUCCESS in 11m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6eaf5400da94e6881c7f7ad910e844e : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc31c6c5e1164289ac1c822cddae154d : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/26da6c5766df4711b2a87bf4235241dd : SUCCESS in 56m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef3c0ccd205945aaa66ec763fe5c51e4 : FAILURE in 1h 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5daf4f26194a463c9a17669e759d086b : SUCCESS in 44m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae3412fdfcf0431dafc1f3c5fa4c0eab : SUCCESS in 8m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26843607946d45db980fe2301414b629 : SUCCESS in 15m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c520496843c40f48f3236aab5911ac9 : SUCCESS in 5m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/64be71392f8946f8a9e4093de6addc9d : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f98445693d9437c9b44eda0df52a31e : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8cb90eb48ccd496f994c6394298598d4 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0de3e28ff4024dc5a889c365b88bd8db : SUCCESS in 42m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96d81bebb19b4a63a8f88b7e8998eadd : FAILURE in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c71662a77e5541dab30d59aabc74a84c : SUCCESS in 1h 12m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/337ed10c269b43d59ca2e6503970f393 : FAILURE in 1h 18m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91985ea3fe5b477494ace8e2d7313306 : SUCCESS in 1h 35m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a94bfce433d4904b6040d92bbc1e52d : SUCCESS in 1h 06m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c2245a7fbfee41ceaf147b998e272d0a : SUCCESS in 1h 19m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a2541bb16135411cbd61fd91330e781d : SUCCESS in 29m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95657af645d14af2b471fd13eb566df4 : FAILURE in 1h 09m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a9cad4c791449fe92851e461c6d1c48 : SUCCESS in 2h 11m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51d013925d194c9db4959723ec1b68ce : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":4},{"id":"426a2efd93ff13bee36dd8b08cc4ae70ad4726ec","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-06-17 11:41:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi finger://vm-css-bionic-2/3636fa3b3c584b16be679e85bbc41729 : POST_FAILURE in 1h 13m 21s\n- storwize-tempest-dsvm-full-fc finger://vm-css-bionic-2/c16883babd6840acafd4c1d48ae3e618 : POST_FAILURE in 1h 15m 56s","accounts_in_message":[],"_revision_number":4},{"id":"a0d7cede21faaac0aa72ebbcf73f5810b7f09497","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-17 13:19:06.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/94ebed1ac9034730942b773ea09c8c1e : POST_FAILURE in 2h 24m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f46114466e8249c4b251cec9ac7ed439 : POST_FAILURE in 2h 42m 32s","accounts_in_message":[],"_revision_number":4},{"id":"933628aec9e448d19c0ace8096f55d791502da61","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 15:48:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerMAX_FC/4c3d73f/DellEMC_PowerMAX_FC/575 : SUCCESS in 3h 10m 10s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerMAX_iSCSI/c78dff7/DellEMC_PowerMAX_iSCSI/562 : SUCCESS in 3h 34m 02s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"e9fcbc87596f32f51a29dca01437db52ad251b53","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-17 20:03:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_SC_FC/4e5b5cd/DellEMC_SC_FC/701 : ABORTED in 3h 00m 01s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_SC_iSCSI/ee1c4ba/DellEMC_SC_iSCSI/722 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"9e066b80e85282dcbaea9eba47a282b5e11ebd49","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-17 23:53:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/806605/4/check/dsvm-tempest-cisco-zonemanager-job/f4d2bcd/ : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":4},{"id":"86b492f7763ddf9b2af1de70f62804ab9482a1fa","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-06-21 09:16:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9b8915b6d0745be8adacdcea1966e706481b78c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-21 09:57:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/504ef52f9d98457480a7bd2f10d997e0 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":4},{"id":"60ef1950405952c14320a33c5dff54ca2433eb5d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-21 10:12:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/4/iscsi/ : SUCCESS in 55m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/4/smb/ : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":4},{"id":"e5452f22e0da8dc0d908371762bc285bf689e6b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 10:16:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e17acb9c1094a1a9ae9f879a0b0acb0 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58fe4790bd6f4af7bbaf6fdecc87c6ce : SUCCESS in 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edbb1df14a4cddb1c18dfcbdb67955e604ff0018","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-21 10:34:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/806605/4/check/dsvm-tempest-cisco-zonemanager-job/ed65b4f/ : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":4},{"id":"53c1656ef4675cc3d2855b09e3d9b39b1247ecd2","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-06-21 10:43:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi finger://vm-css-bionic-2/10fbe2ca293b48c2828672b3fc6297e2 : POST_FAILURE in 1h 12m 58s\n- storwize-tempest-dsvm-full-fc finger://vm-css-bionic-2/7ba9921eafcb4ec6831ba1e94434bc05 : POST_FAILURE in 1h 14m 52s","accounts_in_message":[],"_revision_number":4},{"id":"64027474b5deecacd4847a3c454e7fa640159777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 11:31:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1eb2ef9e00f439492ebcf276ed7ebb6 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4534355f33814d65ada26618fb9c4f9e : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f047265472e841eb9ea1e465d06ab7fe : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e329a89736344932938eb7510f4da305 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4bb56d7ce3a489fa1feecf1805f98c0 : SUCCESS in 16m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/1544361370a7404c99b1def6e8bce0ce : SUCCESS in 55m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de9d1c83858942018b45a842bc98c786 : SUCCESS in 1h 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82e036c918b14bf8a6d41f3303817701 : SUCCESS in 44m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4b915e795ba4e9b8bdb39be03c82fd6 : SUCCESS in 9m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b458e3db68444a2ca03bdd9cb6a3bc88 : SUCCESS in 13m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6440c7d30c874c71936d2268d7ea9ba0 : SUCCESS in 5m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/107bd3a8cccb4480a0c258bbb7f83ec8 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8085b197bb74c87a8e4098b12a7d8db : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09cde3a43af94e5ab8079678475cf589 : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bb1ee8a29fb5484baec6df47bae96cd7 : SUCCESS in 41m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aaef324f93354488af4dcfab3a5b5b11 : FAILURE in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67ef6af4b029454597c50fbe534c44b5 : SUCCESS in 1h 29m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/787ae182a64645b3ae8112c7ebe5f347 : FAILURE in 1h 24m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/976106f2128c4700b7da2c1e9afef256 : SUCCESS in 1h 34m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88413dffb1154c04abb13b06e81f23fd : SUCCESS in 1h 18m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c9d3ea5902243c283c9b515902527f8 : SUCCESS in 1h 31m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/410b0ded82914ce9980590241e019ebe : SUCCESS in 38m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/874c91874ce14cac8a6f2c3ed0eac3dc : FAILURE in 1h 02m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73f901ee46d2424ab6ea8188c30bcb28 : SUCCESS in 2h 12m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c04f536f5a2d4527be0f3f0fcab0eb67 : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":4},{"id":"abecf92fd9713ce71074298bc3975f7260943006","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-21 11:31:19.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"ac0be6c4b23a66b67e31651659d737d80930075c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-21 11:41:42.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8ce1c444cc6444c99b1c43e8616d3eb8 : POST_FAILURE in 1h 44m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/83f62a28c6694a0eafa4ed429d1ce97f : POST_FAILURE in 2h 06m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e98c4a890bc8c010b9be82961025ccdc2c690ade","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 12:48:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerFlex/3d07a33/DellEMC_PowerFlex/874 : SUCCESS in 1h 17m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c5aaead35ca2527744e532bda61ccba627817647","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-21 13:02:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-tempest-plugin-ontap-fcp/43fb348/ : FAILURE in 42m 44s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/35ed3fe/ : FAILURE in 1h 29m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/4/upstream-check/cinder-tempest-plugin-ontap-nfs/d7a817f/ : SUCCESS in 49m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"d98e896e807a7d4583cae5b8e8d92ba28d5a8dfe","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-21 13:23:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_SC_FC/ee979db/DellEMC_SC_FC/720 : SUCCESS in 46m 57s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_SC_iSCSI/0cb2f81/DellEMC_SC_iSCSI/742 : SUCCESS in 1h 52m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"6c1de6776158ae329e45e93f1198ed57f9828a14","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-21 13:27:20.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_Unity_FC/37e1e0e/DellEMC_Unity_FC/984 : SUCCESS in 1h 07m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_Unity_iSCSI/eb088dc/DellEMC_Unity_iSCSI/924 : SUCCESS in 1h 47m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b17e8abd6bb76efb23835753166888f721246415","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-21 13:47:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/806605-4/HPMSA-iSCSI/ : FAILURE in 55:14.97\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/806605-4/HPMSA-FC/ : FAILURE in 49:24.56","accounts_in_message":[],"_revision_number":4},{"id":"53f6b5f03e44fc07dc15945c840bdf11edf8879c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-21 14:13:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/806605-4/PVME-iSCSI/ : FAILURE in 1:08:07\n- PVME-FC https://os-logs.seagate.com/dell-ci/806605-4/PVME-FC/ : FAILURE in 1:04:09","accounts_in_message":[],"_revision_number":4},{"id":"e90ab1f2213c2e03f6c6598e0f49206ede37aa5d","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-21 14:14:22.000000000","message":"Patch Set 4:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/806605-4/Lenovo-iSCSI/ : FAILURE in 1:06:50\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/806605-4/Lenovo-FC/ : FAILURE in 1:06:17","accounts_in_message":[],"_revision_number":4},{"id":"3852dcddb90944cb1b25241ac8668bd8a3437f01","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-21 14:20:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/806605-4/Seagate-iSCSI/ : FAILURE in 1:10:22\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/806605-4/Seagate-FC/ : FAILURE in 1:07:58","accounts_in_message":[],"_revision_number":4},{"id":"9f6a74847230d9ae04af86adb135d2864069caf5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 15:29:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerMAX_FC/dc1438c/DellEMC_PowerMAX_FC/595 : SUCCESS in 3h 37m 18s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/4/check/DellEMC_PowerMAX_iSCSI/f90d269/DellEMC_PowerMAX_iSCSI/582 : SUCCESS in 3h 57m 47s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"9c95d762ccc905d029980ad35bace686f064e332","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-06-21 15:53:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3361c3b474b4ed3159695805f175373d0a776cc1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-21 19:06:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/806605/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2865b07/ : TIMED_OUT in 3h 03m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/806605/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3238a0a/ : TIMED_OUT in 3h 09m 58s","accounts_in_message":[],"_revision_number":4},{"id":"12ff1ab1acc0317289d517499ea56a3f7b42d682","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-22 03:21:59.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/806605/806605/4/check-not-vote/ext-cinder-zuul/e0f0ee9 : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":4},{"id":"3ede46b08626095f48ddb22bd44674a2dcd0643e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-06 15:41:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2d5666bfed52de71300cd14a93c889feaa20b26d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-06 15:58:24.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8ae7800034eeb0abfbab24b4595a0862fcd83d2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-06 16:01:28.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657123114426 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"0fa58698e4343f5c4468b64473485f421511c2cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-06 16:03:10.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"594be6615e46e3936f10ae665d4a9fe364e83f29","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-06 16:03:46.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"425192af55d41aed46f6c8901ae8b7d158a2d9fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-06 16:04:57.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657123399436 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"bc82c76ab5680b67f18c65f1fb733e77c6fdf070","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-06 16:05:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d480df4e8455632651f5f01e049ecaa4ee92d7a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:35:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e761fc421c664f8f95039a647707dce2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c62f0c3d30847f2a53afcb541f86fbb : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ca919aba3bc49b5b2ec00ba90eba49a : SUCCESS in 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dab7b0aa742fb631db40e92652ebd7b317473247","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-06 16:41:42.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"496192ba0b03f85e5692bd265a7d898e6399dab1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-06 17:19:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/806605 : FAILURE in 1h 10m 25s","accounts_in_message":[],"_revision_number":6},{"id":"999af02c8550271c94b5877bafdbd9983f1dc27a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-06 17:33:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/806605/6/check/dsvm-tempest-cisco-zonemanager-job/d39ba8a/ : FAILURE in 53m 26s","accounts_in_message":[],"_revision_number":6},{"id":"946a5cb9fa97ccf69d906eb3fb8ff9b44489e886","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-06 17:35:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/806605/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5c31cef/ : SUCCESS in 1h 23m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/806605/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/47cccd7/ : SUCCESS in 1h 25m 11s\n- pure-devstack-cinder-tempest-nvme-fc-aio http://openstack-logs.purestorage.com/05/806605/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-fc-aio/71c648b/ : FAILURE in 28m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/806605/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/1a538a5/ : FAILURE in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4710c35290e43fddc651b4c534945b8f52939697","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-06 17:44:03.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/806605/6/check/tempest-dsvm-full-focal-py3/8760638/ : SUCCESS in 1h 17m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/806605/6/check/devstack-plugin-ceph-dsvm-focal-py3/a8597f7/ : SUCCESS in 1h 25m 58s","accounts_in_message":[],"_revision_number":6},{"id":"774cdc73451d9e58d9b4e40b06c8dce8de549a2b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-06 18:00:04.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/6/iscsi/ : FAILURE in 1h 56m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/6/smb/ : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":6},{"id":"6ee8dfb11c2b7738ec06589210058a1e9e8ed965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 18:42:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f631f13b212c4b638cf119204948a3b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcfbe72357fa434480f50ed96b7a8685 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/452bd657cfd844e1899f4bb63b5a0fab : SUCCESS in 11m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02246252928e41eba6ba4e6d23e6f20a : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af07f043e5f542229ed4ccc31322a201 : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b30b362d56243ee941e7f3fba1c808d : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/8ad0d67d4a09444fbeb3bcca84814715 : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/537986bdc2c54614bf05b8afe890d488 : SUCCESS in 1h 31m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2ac79f5bb6d4652ad35b45479a09e66 : SUCCESS in 53m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec5e759fae14c0f92e2479adae71168 : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e4fea13412874908a079471d492b0ee4 : SUCCESS in 15m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11eedcd806654bf8b90eba92fc5df063 : SUCCESS in 5m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d6e292a5e704abaaccf36b392f2882c : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73887faf6c5d4d80a2738a20b098e892 : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b68f18641f4471d945e8d6993a5d564 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/484534ed263d4ff2a80716f6134b2167 : SUCCESS in 46m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6bcdc1167cb54d2f90486a1566d31be9 : FAILURE in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6284bd63e26c474dbd2518b4251bd1e6 : SUCCESS in 1h 16m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e1cabfea1b7471d850a934a27e559e6 : FAILURE in 1h 34m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a2c04c9f87e4f659305829753a2f1f4 : SUCCESS in 1h 53m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/361b671fd5174f53b24b38ab0273acec : SUCCESS in 1h 08m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aba0076fa08440c7af47ab59b0a52474 : SUCCESS in 1h 18m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/448f398cefd74ee5b2198f3ac2f2c232 : SUCCESS in 46m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9e2251264e64430881683fb9c3cafe50 : SUCCESS in 1h 18m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6487c922a57549b1a3013e76c1d8d9aa : SUCCESS in 2h 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7060d168b8ca49c3b18d2124ddca9725 : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":6},{"id":"eb9816f093394d47be721247008d0fa926745e9f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-06 19:22:36.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/05/806605/6/check/DellEMC_PowerStore_NFS/267aecc/DellEMC_PowerStore_NFS/77 : FAILURE in 40m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"0893548efe6b7019d29e66ee1f3e1274c3ffa587","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-06 19:33:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/806605/6/check/iscsi-driver-nimble/a253c6c/a253c6c.html : FAILURE in 1h 32m 42s","accounts_in_message":[],"_revision_number":6},{"id":"037d3cc8057b5416e24851cb96922a09e315f500","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-06 20:33:36.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/05/806605/6/check/DellEMC_PowerMAX_FC/647fd4f/DellEMC_PowerMAX_FC/682 : FAILURE in 38m 05s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/6/check/DellEMC_PowerMAX_iSCSI/a085169/DellEMC_PowerMAX_iSCSI/671 : FAILURE in 50m 45s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"db1ad470b2cb7d13d6729da162c9a4fed6937670","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-06 20:34:21.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-806605-5/ns5_nfs : SUCCESS in 3h 26m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-806605-5/ns5_iscsi : SUCCESS in 4h 7m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"96924cc6ac8dad3fb2208116ae5fd6fbddf8c17b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-06 20:55:54.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-806605-6/ns5_nfs : SUCCESS in 4h 4m 13s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-806605-6/ns5_iscsi : SUCCESS in 4h 6m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"cda9f182e533789f8d2ffd45c15d50c4a41a044c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-06 22:28:52.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/806605/6/check/DellEMC_Unity_FC/de9bb8a/DellEMC_Unity_FC/1104 : FAILURE in 1h 09m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/6/check/DellEMC_Unity_iSCSI/0e6fa33/DellEMC_Unity_iSCSI/1036 : SUCCESS in 1h 52m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"74841f3ebdb33184311d8ac4bede84d1832b66af","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-07 08:22:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"142ca5dd65d318305b4145b48b39117728793155","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-07 08:36:04.000000000","message":"Patch Set 6:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":6},{"id":"1542de2e6ef2c8210b2b7e369a96c1bfd9586f97","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-07 08:36:12.000000000","message":"Patch Set 5:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":5},{"id":"cf048c5365993add8b62a77ba09a1c32fea60861","author":{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"},"date":"2022-07-12 12:06:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5ae289375b02765c78cec34e2366a26d88b81f36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-12 13:31:40.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/6/check/DellEMC_PowerFlex/3f331ab/DellEMC_PowerFlex/973 : SUCCESS in 1h 25m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"8c7099bceb4eb0f00c3a9a8bdb62c13050e43e31","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-18 20:49:22.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/6/20220719045212/NEC-V-ISCSI : FAILURE in 51m 47s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/6/20220719045212/NEC-V-FC : FAILURE in 57m 06s","accounts_in_message":[],"_revision_number":6},{"id":"742f832290cc5524c474fe9e64e0cd7f73475a94","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-23 07:45:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/6/20220723131449/NEC-FC : SUCCESS in 3h 27m 25s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/6/20220723131449/NEC-ISCSI : SUCCESS in 3h 30m 59s","accounts_in_message":[],"_revision_number":6},{"id":"c5d927d362c8539f56281baff54452b8e43fc407","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-05 09:10:05.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6eb3dbe1e6325ba87c79142a9aee3dcd4f2e7875","author":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"date":"2022-08-05 09:37:09.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"409084af63e9216f041dd8ceb0897226d58808ba","author":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"date":"2022-08-05 09:41:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7c0174a1547a8f052a9b231f483571e0d0a4d841","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-08-11 14:00:33.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b59c9bcb3de8717d7b27ec1908d322e6b91fcac0","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-26 06:04:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fe5e0d26e11689790b99ba55d5ba6e9aeda0b6a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-08-26 06:53:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5af2a0bf0ee0afcd05c8f271fd79568137706655","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-08-26 06:55:01.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9fdaf2312344221240b6f4f14cc8bff162ff98b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-26 07:11:31.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661496800435 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"10ba8bda2cc6d3a6e87a8e0c66fcb0944959833c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-26 07:16:10.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/806605/7/check/dsvm-tempest-cisco-zonemanager-job/9261f83/ : POST_FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":7},{"id":"0a34fe713307daa6759df3974b7926fc3e55b3c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 07:22:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e045a4f465845d9a29d46eb8e9c647f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc9fc26d1d964eb0a723203af3d33d54 : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8789ad190fd84fdf9c7b4815302b8b48 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e08bf304c9880c206f71dc6b2c678d5fe820cf90","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-26 07:58:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/806605/7/check/iscsi-driver-alletra-6k/986349e/986349e.html : FAILURE in 1h 03m 57s","accounts_in_message":[],"_revision_number":7},{"id":"4252e58e39133c410dd6f8ec7aee8754e16d307c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-26 08:06:14.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/806605/7/iscsi/ : SUCCESS in 1h 11m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/806605/7/smb/ : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":7},{"id":"331e41664cb45837b7cd8846ce083469a2149bd6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-26 08:24:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/806605/7/check/fujitsu-eternus-dx-iscsi/54c3b65 : FAILURE in 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/806605/7/check/fujitsu-eternus-dx-fc/6a14063 : SUCCESS in 1h 30m 40s","accounts_in_message":[],"_revision_number":7},{"id":"012e9e0552ccd85fbfb64b38c4e828edfeb22a70","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-26 08:29:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/806605 : SUCCESS in 1h 34m 22s","accounts_in_message":[],"_revision_number":7},{"id":"e8b34ebdd1e4fa78bb5525e694e6214169e4df13","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-26 08:30:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/806605/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/80c76c1/ : SUCCESS in 1h 29m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/806605/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5a55ed6/ : SUCCESS in 1h 35m 41s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/806605/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/829f0dc/ : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":7},{"id":"661c0bb04a76a1ce9fb13ca69be0eb5334c00d5b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-08-26 08:44:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/806605/7/ : SUCCESS in 1h 43m 31s","accounts_in_message":[],"_revision_number":7},{"id":"49b4115655658ea0de34a171f9e3641261425c9c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-26 09:08:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/806605 : FAILURE in 2h 13m 19s","accounts_in_message":[],"_revision_number":7},{"id":"3a025a344118cf7421e923e28ff5dfa470c6cb55","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-26 09:12:17.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/806605/7/check/tempest-dsvm-full-focal-py3/9ba29f1/ : SUCCESS in 1h 48m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/806605/7/check/devstack-plugin-ceph-dsvm-focal-py3/28b5a93/ : SUCCESS in 2h 13m 53s","accounts_in_message":[],"_revision_number":7},{"id":"3b436f27cd8c870e85e73a51f0b440335a8a3ea0","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-26 09:16:15.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"63f13913c43c0dee96bcfe39484689727a5c0ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:26:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c2964ff7134aacb38886c93e6f5109\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5ef48739eb43b382dae3e65b0da38d : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f838ffbeb4924d3b90ac387931c7c348 : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e8e34dd9df84fa08b1543472202b6a5 : SUCCESS in 11m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d14af4f36bb548b48aa627908a3cf7aa : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b20ede484e04e70b81a86c3c902fc70 : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/38f2a95db74c46f5bcfbd82d150e521b : SUCCESS in 1h 04m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d16d0102cf4d4fbbbdf3732181620c0b : SUCCESS in 1h 30m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a95951417db4a5bb9e64433cab89074 : SUCCESS in 1h 04m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9731fa079baa4c45b66e560b662bd843 : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ea1cc17ba884d56a4d266bea503d16c : SUCCESS in 16m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1cbc2504ae384098999360e8608b6ef2 : SUCCESS in 8m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b912316bce14cc2b43fa9a7d739e38a : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/347bc80cc42544689e5e466ffffdd1e2 : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6f846b12995f4d359278b2a4123c44ea : SUCCESS in 9m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d16bb1c89ac144dd96ae38e6c53217d9 : SUCCESS in 38m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63776b37d6fc4836972fd005a5b15a92 : FAILURE in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/908afe8759cc4270a96cae79ffde5991 : SUCCESS in 1h 24m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e69d0fc7c2ce4dbea93b8d9d1d536010 : FAILURE in 1h 38m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bafbfd6ae2c44185b08c81e5e1b48faa : SUCCESS in 1h 43m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9cae61456c274e51bc0e25bde3295e9d : SUCCESS in 1h 19m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c68eab57b1044ab938370d282ee3f08 : SUCCESS in 1h 28m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d6debd11e9704d57890eebcf01eb4305 : SUCCESS in 51m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f88e0fb6bbd4de8a6c1c766802f0805 : SUCCESS in 1h 24m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d748ce5b6444efc81224002ee897563 : SUCCESS in 2h 23m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eec043e631947d88bedf4df242d8516 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":7},{"id":"3e1a5bb9ada2501249e7e4841bebb69c57279b25","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-26 10:28:02.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-806605-7/ns5_nfs : SUCCESS in 3h 32m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-806605-7/ns5_iscsi : SUCCESS in 3h 34m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"48eecdb82555f7e1b2a864ab2f82b78c7ffeb260","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-26 10:58:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/7/upstream-check/cinder-tempest-plugin-ontap-fcp/4b8164a/ : RETRY_LIMIT in 8m 02s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/7692fa0/ : SUCCESS in 43m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/7/upstream-check/cinder-tempest-plugin-ontap-nfs/4d70be2/ : SUCCESS in 43m 52s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/806605/7/upstream-check/cinder-tempest-plugin-solidfire-iscsi/8548a36/ : SUCCESS in 1h 30m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"0734ac4a5972237785c8faa1290ff642d18bf081","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-08-26 12:45:58.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"062b47baff071c61172732309ac9e6f2bddac1b0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-26 13:14:11.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_PowerFlex/481b02a/DellEMC_PowerFlex/1227 : ABORTED in 2h 00m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b876071bd8963a72a1b000a2eebd87e7c250bfff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-26 14:17:14.000000000","message":"Patch Set 7:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/501","accounts_in_message":[],"_revision_number":7},{"id":"ee920014d62e16b7e47a4857f24ca643dc54e8b1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-27 02:11:43.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_PowerStore_FC/35e9594/DellEMC_PowerStore_FC/796 : SUCCESS in 1h 20m 31s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_PowerStore_NFS/0fa1199/DellEMC_PowerStore_NFS/335 : FAILURE in 56m 51s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_PowerStore_NVMeTCP/4d20c9c/DellEMC_PowerStore_NVMeTCP/210 : FAILURE in 1h 51m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"3c05838477e0085264936b2137923370d9856ab7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-27 14:30:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/806605-7/HPMSA-iSCSI/ : SUCCESS in 1:05:47\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/806605-7/HPMSA-FC/ : SUCCESS in 58:23.33","accounts_in_message":[],"_revision_number":7},{"id":"17e679f6c0d14321e7523d82ec580c6fef0b5283","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-28 05:23:46.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/806605-7/PVME-iSCSI/ : SUCCESS in 1:08:21\n- PVME-FC https://os-logs.seagate.com/dell-ci/806605-7/PVME-FC/ : SUCCESS in 1:05:19","accounts_in_message":[],"_revision_number":7},{"id":"df376bf2e342aa057bbedff1d7bc196755ca35fe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-28 07:05:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_Unity_FC/d19e5df/DellEMC_Unity_FC/1581 : FAILURE in 15m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_Unity_iSCSI/22d3dfb/DellEMC_Unity_iSCSI/1450 : SUCCESS in 2h 01m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"8f1c9bba2238b91e9dd1090a4ede85aee2eac248","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-28 09:12:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/806605-7/Lenovo-iSCSI/ : SUCCESS in 1:08:18\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/806605-7/Lenovo-FC/ : SUCCESS in 1:05:37","accounts_in_message":[],"_revision_number":7},{"id":"e25a574b77f7855408f06ee1e7cf87d3ab4d6433","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-28 11:51:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/806605-7/Seagate-iSCSI/ : SUCCESS in 1:08:12\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/806605-7/Seagate-FC/ : SUCCESS in 1:07:04","accounts_in_message":[],"_revision_number":7},{"id":"141a009d158c41f741762f9f175cd9bd811bc62c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:35:28.000000000","message":"Patch Set 7:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":7},{"id":"c64d0a73ac47773aba5de986d5bf49c99e00a201","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-31 19:46:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/7/20220901011314/NEC-FC : SUCCESS in 3h 26m 53s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/7/20220901011314/NEC-ISCSI : SUCCESS in 3h 32m 44s","accounts_in_message":[],"_revision_number":7},{"id":"4f1da6f32560db837827eb0a08dab7c88b6f77e9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-05 10:11:21.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bdec28cd6bcb928d629a36720feac08dc75dac21","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-07 00:23:17.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/7/20220907082759/NEC-V-ISCSI : FAILURE in 53m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/806605/7/20220907082759/NEC-V-FC : FAILURE in 55m 14s","accounts_in_message":[],"_revision_number":7},{"id":"1fe87aadcb776425a773c20e0abf01c046cb7337","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-09-14 12:22:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4b1e21a4d110f0ff45c6ec39a8757719fc7ac16f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-14 14:22:26.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_PowerFlex/2aed82d/DellEMC_PowerFlex/1320 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"97eda3481e8defcc5c5b2a4e4f68d0984b6511e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-15 04:47:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/47eaeb6afdd14562890298a9a3398e82 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":7},{"id":"ed3d041fff4fc02513f203787c98bb4c630cd50f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-15 13:23:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f38f03ce9c5cdff269ea833d7fed5568476b43b1","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-09-15 14:08:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"be533396b12591edea0716f492a63b6b9e609f47","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-15 15:57:22.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/05/806605/7/check/DellEMC_PowerFlex/3e3e711/DellEMC_PowerFlex/1325 : FAILURE in 1h 48m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"cd759cee368321823e9c6482fb7834a055813e38","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-15 18:13:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d58c8e6ceb558baaf4db07a7b140367aa8ff46e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 18:13:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"12e51f0ee2811be0f6b8fce46c4678d33519846e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 19:25:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b28323f7e6e94460b2afbc13b9842147\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a812300de24bc79c1cfc2f0c3b9b5f : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5792c487a46f4e6ab52ab3ab2565f21a : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/073f91686a124bd8ae1017088aba41e0 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfe1651a4394d94b0b7a0954fe58270 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c045f77c7d974a208299bed04bc9c650 : SUCCESS in 47m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/762d6d655eb9462b9801b5168e6b6f47 : SUCCESS in 1h 06m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4105052bfdf4635a9808c1ada57f439 : SUCCESS in 1h 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fde601c9eda4a3888ea5b9b4b37a9af : SUCCESS in 6m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/01a41530c77c4a01aa2a56fc75009d4c : SUCCESS in 55m 00s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5876dd514a745e0af72cafa7e7b59c1 : SUCCESS in 1h 01m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a24adea356b4c16b7eb612efe140c1d : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":7},{"id":"bfe28b2f289496310e8842aa84c23e0d226eaf70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 19:25:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"0acf589f018adf6de75bc085c46b5f364b349f0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 19:26:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3160dbef6e748a5ab6319e82b6d2194\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4f57efca27405eba8bda0e35167c93 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fd6292ee61641118038df28e1cee661 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":7},{"id":"0ffa57b21fff7b430ad26ec8951badac54b1fef5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-16 18:15:14.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-806605-7/ns5_nfs : FAILURE in 28h 11m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-806605-7/ns5_iscsi : FAILURE in 25h 6m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"05de32ece711eb101041efd153852121cb2e5fd8","revisions":{"26716c19f1502db1e123fb58002b55030d6cc3ec":{"kind":"REWORK","_number":1,"created":"2021-08-30 14:56:56.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/05/806605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/1"}}},"commit":{"parents":[{"commit":"6901d2337c05fa629c8170991aa85afb7432eca0","subject":"Merge \"[Pure Storage] Resolve missing provider_id issue (PowerVC)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6901d2337c05fa629c8170991aa85afb7432eca0"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26716c19f1502db1e123fb58002b55030d6cc3ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26716c19f1502db1e123fb58002b55030d6cc3ec"}]},"branch":"refs/heads/master"},"86738f2614601d1ffbb963e330e831329bbf9559":{"kind":"REWORK","_number":2,"created":"2021-08-30 15:25:24.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/05/806605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/2"}}},"commit":{"parents":[{"commit":"6901d2337c05fa629c8170991aa85afb7432eca0","subject":"Merge \"[Pure Storage] Resolve missing provider_id issue (PowerVC)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6901d2337c05fa629c8170991aa85afb7432eca0"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 15:25:04.000000000","tz":180},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86738f2614601d1ffbb963e330e831329bbf9559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86738f2614601d1ffbb963e330e831329bbf9559"}]},"branch":"refs/heads/master"},"9e21b69e7e1410c5aaa338bc0e10366a1b3b16df":{"kind":"REWORK","_number":3,"created":"2021-08-31 12:27:32.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/05/806605/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/3"}}},"commit":{"parents":[{"commit":"6901d2337c05fa629c8170991aa85afb7432eca0","subject":"Merge \"[Pure Storage] Resolve missing provider_id issue (PowerVC)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6901d2337c05fa629c8170991aa85afb7432eca0"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-31 12:27:04.000000000","tz":180},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e21b69e7e1410c5aaa338bc0e10366a1b3b16df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e21b69e7e1410c5aaa338bc0e10366a1b3b16df"}]},"branch":"refs/heads/master"},"4133608e15a82494a454b622e87695fe58a29444":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-10-01 10:07:16.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/05/806605/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/4"}}},"commit":{"parents":[{"commit":"555ef558036d1568402c880f1c4a733cd0c587c6","subject":"Merge \"HPE 3PAR: Reuse existing session\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/555ef558036d1568402c880f1c4a733cd0c587c6"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"vladislav.belogrudov@dell.com","date":"2021-10-01 10:07:16.000000000","tz":0},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4133608e15a82494a454b622e87695fe58a29444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4133608e15a82494a454b622e87695fe58a29444"}]},"branch":"refs/heads/master","description":"Rebase"},"aaa22b8a1c3fdd61f0b8a998b6dc2904f19650b8":{"kind":"REWORK","_number":5,"created":"2022-07-06 15:58:24.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/05/806605/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/5"}}},"commit":{"parents":[{"commit":"555ef558036d1568402c880f1c4a733cd0c587c6","subject":"Merge \"HPE 3PAR: Reuse existing session\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/555ef558036d1568402c880f1c4a733cd0c587c6"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-06 15:58:06.000000000","tz":0},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaa22b8a1c3fdd61f0b8a998b6dc2904f19650b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaa22b8a1c3fdd61f0b8a998b6dc2904f19650b8"}]},"branch":"refs/heads/master"},"c151140ec0c461fdb6fcdac80eb9294286cceb13":{"kind":"REWORK","_number":6,"created":"2022-07-06 16:03:10.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/05/806605/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/6"}}},"commit":{"parents":[{"commit":"555ef558036d1568402c880f1c4a733cd0c587c6","subject":"Merge \"HPE 3PAR: Reuse existing session\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/555ef558036d1568402c880f1c4a733cd0c587c6"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-06 16:01:49.000000000","tz":0},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c151140ec0c461fdb6fcdac80eb9294286cceb13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c151140ec0c461fdb6fcdac80eb9294286cceb13"}]},"branch":"refs/heads/master"},"05de32ece711eb101041efd153852121cb2e5fd8":{"kind":"REWORK","_number":7,"created":"2022-08-26 06:53:09.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/05/806605/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/806605/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/806605/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/806605/7"}}},"commit":{"parents":[{"commit":"555ef558036d1568402c880f1c4a733cd0c587c6","subject":"Merge \"HPE 3PAR: Reuse existing session\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/555ef558036d1568402c880f1c4a733cd0c587c6"}]}],"author":{"name":"Vladislav Belogrudov","email":"vlad.belogrudov@gmail.com","date":"2021-08-30 14:52:31.000000000","tz":180},"committer":{"name":"happystacker","email":"jeanpierre.roquesalane@dell.com","date":"2022-08-26 06:53:07.000000000","tz":120},"subject":"Fix volume caching in PowerFlex driver","message":"Fix volume caching in PowerFlex driver\n\nVolume caching stops working after 126th volume created from\nthe same image due to snapshot limit per volume in PowerFlex.\nThe driver needs to throw exception \"SnapshotLimitReached\" to\ninvalidate cache and start over with new volume snapshots.\n\nCloses-Bug: #1942095\nChange-Id: Ic61f23c87de5dad52c06061849e3db970cee4327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05de32ece711eb101041efd153852121cb2e5fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05de32ece711eb101041efd153852121cb2e5fd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
