)]}'
{"id":"openstack%2Fcinder~810000","triplet_id":"openstack%2Fcinder~master~I34f8d50cab626526cfd2416f605cc0f2a9325e88","project":"openstack/cinder","branch":"master","topic":"drop-sqlalchemy-migrate","hashtags":[],"change_id":"I34f8d50cab626526cfd2416f605cc0f2a9325e88","subject":"Fix cinder-manage db version","status":"MERGED","created":"2021-09-20 11:26:39.000000000","updated":"2021-09-23 18:31:06.000000000","submitted":"2021-09-23 18:17:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810000-drop-sqlalchemy-migrate","meta_rev_id":"4ca9e329eabd91254d0220c8bb0b30ec1b5fbde4","_number":810000,"virtual_id_number":810000,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 18:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-09-23 13:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-23 14:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-23 14:18:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-09-21 22:11:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 11:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 12:20:37.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 12:28:40.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-09-20 12:29:20.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 12:30:15.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-09-20 13:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 14:15:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-09-20 14:22:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 14:38:43.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 15:25:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-21 17:58:25.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-09-21 22:11:42.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-09-22 12:49:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-22 15:31:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-09-22 16:01:17.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-09-22 16:09:20.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-22 16:17:09.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-09-22 16:28:13.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-09-22 16:34:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 14:18:26.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"8b87d90f5bcb615bac24748475f4cd4e57b1bb5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-09-20 11:26:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65eb116f22a49ebbc1e9a82e35b6f75cebf5740f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91e99e8583874c78af0781526b0181dc : FAILURE in 20m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da1e7276bd3c4e66a54981e7aa371c46 : FAILURE in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ceac260e687c707e79734ad727505e47d07ace26","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-09-20 12:20:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/810000/1/ : FAILURE in 49m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c4707742c5550905f95a6e55162fed11869f3e17","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-20 12:28:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/810000/1/check/dsvm-tempest-cisco-zonemanager-job/7d86b6a/ : FAILURE in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e7fe30e4c84d046f5cea4da19aaca3f2bd8c6dc9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-20 12:29:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/00/810000/1/check/iscsi-driver-alletra9000/b751d3f/b751d3f.html : FAILURE in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f3719529b688ed4e7cefd6911bbbf90ae9eab591","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-20 12:30:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810000/1/iscsi/ : SUCCESS in 48m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/810000/1/smb/ : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"25aa2908834327d311fa37ebea9e1a9d80d951ef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-20 13:08:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/810000/1/check/iscsi-driver-alletra-6k/158ead6/158ead6.html : FAILURE in 1h 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c92d301f8ae532690480727d29a96d2dda54c085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 13:22:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d135ef366a124db7a326194abeea23e8 : FAILURE in 15m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb56d87f74e04aad82ce6895e52d7060 : FAILURE in 15m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33f92a9cf15d4b899efac440cbba1416 : FAILURE in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c9aa4573594fed9d174a3138b76274 : FAILURE in 11m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4786851b042b4cedb63a797e6466b556 : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/89ecd10beadb4b5099678592a016dd42 : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/733a18a5729448e5bb8f011421ed9a16 : SUCCESS in 1h 04m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f8f6b9b4ef6497d9af46c7e41d41749 : FAILURE in 12m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d8c3c0543a504cbf96178ef15b72bb04 : SUCCESS in 6m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/154c9472904f4530b961e0ecdd61fc66 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9f12c31ee9d44e7b8ce9a80fbb33714 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b19ea5d48e5e42b9a89f1acf065bc83c : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7b4aa2b47934d4cacad45e94a91b3f9 : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19bf49b483c540ed9fb74174f110db84 : FAILURE in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5aa374e27734bb2a0d8b313eb9682d2 : SUCCESS in 1h 15m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f2130ad9da543bb9758016510c9381e : SUCCESS in 1h 21m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85c65e2eaf5744c28845194d88d156e1 : SUCCESS in 1h 34m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1fcfebcfd5f4646a87723546d92f988 : SUCCESS in 1h 14m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/37fd91bb37e646a7b7050c363f4d4e5c : SUCCESS in 1h 27m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0193359c0b3f4832bb400f6191676816 : SUCCESS in 38m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a3bc8fceab2e495e8267d6220167e093 : SUCCESS in 1h 13m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7634226581b44e4aff45d5866510b41 : SUCCESS in 1h 53m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80c9431ea11b46f5b98cb56c7b14f5c6 : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31777f99a9f4455e986d6b39d94ee64b : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0bca81a3d349a8e1d5947d75e5dc0ca68862f4ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-20 13:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/810000/1/check/iscsi-driver-nimble/ccc1d17/ccc1d17.html : SUCCESS in 2h 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"77948635dec025a4a34e2c18d2327de788496a07","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-20 14:15:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/810000/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/572a3ce/ : SUCCESS in 2h 18m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/810000/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5030ebe/ : FAILURE in 2h 47m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fdd33a3493f02bd3392df6378a9c1aacdd827460","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-09-20 14:22:41.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/1/check/tempest-dsvm-full-focal-py3/8ff64d4/ : FAILURE in 1h 34m 16s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/1/check/devstack-plugin-ceph-dsvm-focal-py3/1d68406/ : SUCCESS in 2h 51m 05s","accounts_in_message":[],"_revision_number":1},{"id":"938c49fa4561474a7e5418f3c4dd35f427942de7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-09-20 14:38:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/00/810000/1/check/DellEMC_XtremIO_FC/ebfb71e/DellEMC_XtremIO_FC/1022 : SUCCESS in 1h 09m 38s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/00/810000/1/check/DellEMC_XtremIO_iSCSI/c4963e9/DellEMC_XtremIO_iSCSI/909 : FAILURE in 2h 42m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"b4bbd9a46029dfc911802cc4a6fde934bedc6873","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-20 15:25:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-810000-1/ns5_nfs : SUCCESS in 3h 56m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-810000-1/ns5_iscsi : SUCCESS in 3h 58m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"29f903a5109b18f8537cfc5f0b08f39c0d769a9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-09-21 16:20:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b04fe4325e7dced527a90d2409a3f3e5c0abee8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 16:47:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/810000/2/check/iscsi-driver-alletra-6k/6e5d104/6e5d104.html : FAILURE in 25m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4559ab14c06b6d07528be64244dcd092ef119ba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 16:59:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6de0ef9825264dd3ab09bda7d96be5ec : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ebacb7bdc5c410b918603cafd642881 : SUCCESS in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df8a94164fea6d9ac98b98d1ed47f76e1a173de9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 17:23:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/00/810000/2/check/iscsi-driver-alletra9000/8a3c595/8a3c595.html : FAILURE in 1h 00m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6bcda304dd2ddebe9e883dc016aa6194ae8fdd8a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 17:29:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/810000/2/check/iscsi-driver-nimble/c6b08e3/c6b08e3.html : FAILURE in 25m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f86f8e67248ac062e41d5f2e8221183395220c2b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-21 17:51:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810000/2/iscsi/ : SUCCESS in 1h 27m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/810000/2/smb/ : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":2},{"id":"fd0664e0ecb4956ad05e05073acc10c70544844d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-09-21 18:20:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/810000/2/ : FAILURE in 55m 28s","accounts_in_message":[],"_revision_number":2},{"id":"cbd6770bfb46f9da9b4a613e3b37ce46cca8a9dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 18:21:54.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ffa0973ba7042b2bb64264dcf237528 : FAILURE in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86a29b8b74e44073a29aca1f1a2456f4 : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dec6a8b6de7483f89a359cc7fe65b39 : SUCCESS in 13m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d25647b4238c4b97bb761f663fe87f96 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03fb783bbc9a425ba7c824d75bbe142f : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/582db9a79d8f4db8bf9951520e9277d3 : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b58888766a984b76888aad7c90fe426c : SUCCESS in 1h 08m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3a1355443b034c3cb4e9ca6966e10ba3 : SUCCESS in 15m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b1202120263049aabb5f4ca80e61adcc : SUCCESS in 6m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c02643851e7940c2beb858401211c59b : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e494943873374705b6dbf74352be43ba : SUCCESS in 7m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9cb555fcc3654375a4e656276a00a49e : SUCCESS in 6m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8eaa0fca9e9b430a9681a3d449bd9711 : SUCCESS in 41m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a387e58d97584151b82f9fc670df8691 : FAILURE in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1c6d49f5f5b84fdc88552f29790773a2 : SUCCESS in 1h 25m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/68b61e3c2e9045a29969a4f7540778a8 : SUCCESS in 1h 20m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/824d657d7642440aa3058d1488b0a2ff : SUCCESS in 1h 21m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d87c5a44c0b4e3b827884ddb4631f8d : SUCCESS in 1h 15m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/702c106493ef407d8310f07a90ba05e2 : SUCCESS in 1h 31m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3325352c1ee945dc90665382b9a340c3 : SUCCESS in 45m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3017bb16e400411096af561a863031c3 : SUCCESS in 1h 13m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/017717fb329e404599c0e8ebcf7e2607 : SUCCESS in 1h 55m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68319771dfde469aa9de0643c34be58c : SUCCESS in 45m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/358c86523cbf4ab3b573ea2ad86bf86c : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":2},{"id":"fa6ab4d1d716f502e54efb0031df7017d3529de4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-21 18:39:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/810000/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5e98b33/ : SUCCESS in 2h 17m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/810000/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f8eebf3/ : SUCCESS in 1h 24m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d66c5440804484a4d80eba34effbc7ff4c8d2d8a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-09-21 18:59:43.000000000","message":"Patch Set 2: Code-Review-1\n\npep8 failure","accounts_in_message":[],"_revision_number":2},{"id":"c6447c8bc4fad21fa9e1c402bea7260f07c760ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 19:23:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/00/810000/2/check/iscsi-driver-primera/1c27aa8/1c27aa8.html : TIMED_OUT in 3h 00m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f273ac59cf863dba73a49c6bf264b2dbe2d7f1c0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-09-21 19:44:34.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/2/check/tempest-dsvm-full-focal-py3/80c8f1c/ : SUCCESS in 2h 33m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/2/check/devstack-plugin-ceph-dsvm-focal-py3/34cdc61/ : TIMED_OUT in 3h 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"305c6a63ca238beef2d80760ba251fe1b906b158","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-21 20:17:29.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-810000-2/ns5_nfs : SUCCESS in 3h 56m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-810000-2/ns5_iscsi : SUCCESS in 3h 56m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"84f9e7e6c3444feaab2634b64b331e873a18eacc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-21 21:04:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/810000/2/check/dsvm-tempest-cisco-zonemanager-job/1368c01/ : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e163b82f1893097e564e699a60aa3a0283242f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-21 22:11:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"260e0dbad19e0ad74a97fa60efa962d66e8d3267","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-21 22:11:42.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"f56dd94603ffe2416569973ceb463e7740ef9a65","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 22:37:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/810000/3/check/iscsi-driver-nimble/8ac5465/8ac5465.html : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":3},{"id":"d759fd21999616106baa65cb0cc16837a8128a84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 23:14:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/00/810000/3/check/iscsi-driver-alletra9000/a9eded9/a9eded9.html : FAILURE in 1h 01m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9af0f7d0d6a292045f9c8ea588403dc82380bd76","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-21 23:21:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/00/810000/3/check/iscsi-driver-primera/fc3fcd6/fc3fcd6.html : FAILURE in 1h 08m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e390849ea3321910004c3c3d0681de4dbc858d71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 23:27:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a4bfce2475246a4b83c7329d46416c8 : SUCCESS in 14m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78422af3cef54ebcaa1f073e4a467b77 : SUCCESS in 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43a6f4543797acf0be6405a63338cd3392bb6ebc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-09-21 23:43:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/810000/3/ : FAILURE in 41m 44s","accounts_in_message":[],"_revision_number":3},{"id":"222e808548a92a396800cdd1036906d140a857de","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-21 23:53:01.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810000/3/iscsi/ : FAILURE in 1h 41m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/810000/3/smb/ : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":3},{"id":"09ad7380b47ca0d050feb7b53f5ab9130b2b0511","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-22 00:01:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/810000/3/check/iscsi-driver-alletra-6k/3b61f6b/3b61f6b.html : FAILURE in 1h 43m 35s","accounts_in_message":[],"_revision_number":3},{"id":"42c79ab1bda8ba5a6bcc8de4f7772cf9473bb969","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-22 00:20:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/810000/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5bd11cb/ : SUCCESS in 2h 08m 41s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/810000/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2ed8e59/ : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2f40871df12cdbff5376878f077f0c581849982a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-22 00:23:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/810000/3/check/dsvm-tempest-cisco-zonemanager-job/d3e86a2/ : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0bce2a018682a9a8e8a17b2e47776a34969b2afa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-09-22 00:46:10.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/3/check/devstack-plugin-ceph-dsvm-focal-py3/63f0d88/ : SUCCESS in 2h 28m 52s","accounts_in_message":[],"_revision_number":3},{"id":"6fd7bbc6c29abcea8814f033248374eb7581521a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 01:02:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae8c17f612b4cf2ad19cf94401a6de8 : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec8b443415644beb7cd3a51fbdb5e0f : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c89827342b41e181a04acf1f0ffc66 : SUCCESS in 12m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e4ddc8ee78d4b4eb582fc64214d26a1 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614d6477bbc34bef864c7a98f518b839 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/7762591d022e4df193f5d2af8e5bf8ba : SUCCESS in 1h 04m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd1e152ffda843cba940737ee3558a1e : FAILURE in 1h 04m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dbc5e57cc31a427780157781ea2dc6dd : SUCCESS in 15m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cca39e7672444ed4bc1bea099d07009b : SUCCESS in 7m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7cb799ebee30449e8c9976962663bbf5 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f008f34f1954f0bb5640878febf5ce9 : SUCCESS in 9m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb6c474d50764a71a57f5a9dd5f7ff88 : SUCCESS in 7m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/24013f18596c43e1a7870da6efc83cd0 : SUCCESS in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bae85ff5c42d4852af5e9ca67cb9e792 : FAILURE in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fb80f28053e84725a3865c7a9d53c0f6 : SUCCESS in 1h 15m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/09648038047347a89e0885f2ebadb1fa : SUCCESS in 1h 30m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9272444767bb434db2048fc41a4b1253 : SUCCESS in 1h 44m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b1ebf98dcdc4ee9b4bd79266513cfd3 : SUCCESS in 1h 12m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21f1338728b44b4facf9c77aa9f08c91 : SUCCESS in 1h 12m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54fa8e3c315f424792bbf3c23aaa1221 : SUCCESS in 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77602e62e672456889977d1d69beea02 : SUCCESS in 1h 13m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/63adccf5291a49718d578cc5fb5fd382 : SUCCESS in 2h 00m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bacfbc56317c4d06b70ec1fb244d0503 : SUCCESS in 52m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf4d6437333f4a34adb41254e2e09bde : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a210c8085affcc730df84c0d3e3beb4e355754e3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-22 02:08:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-810000-3/ns5_nfs : SUCCESS in 3h 55m 41s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-810000-3/ns5_iscsi : SUCCESS in 3h 56m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"5cbd9c6baa1d041f683a18c913a9cc80e5233dd6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-22 12:13:44.000000000","message":"Patch Set 3:\n\nrecheck tempest-integrated-storage : failure due to a 503","accounts_in_message":[],"_revision_number":3},{"id":"862f420b93799896e4db397045d5f87b1c8e5375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 12:44:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/394c4ca80a2e4a3cafd00cc9cf7f3730 : SUCCESS in 15m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66222d0aee5e4169872b8b73880efe2a : SUCCESS in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b972bc7618f86a920ddaca3a110e4982f74135f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-22 12:49:01.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/810000/810000/3/check-not-vote/ext-cinder-zuul/643f5c0 : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a7ec3f337325b0c8c0789c0a3aa6ae21db7751c8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-22 13:25:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810000/3/iscsi/ : SUCCESS in 1h 10m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/810000/3/smb/ : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":3},{"id":"1fa2095da6d825139ece70183f2df309a5e47adb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-22 13:28:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/810000/3/check/dsvm-tempest-cisco-zonemanager-job/7740c5f/ : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e64f643df5716a076827ca7ec65407264cc5dc81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:32:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/837c7ec81e7040b28b5f65a327dffad4 : SUCCESS in 10m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbc9a2a2c207486cbc53ca1b4a0e21f3 : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5d95b2588264d5a81ade04535535def : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/017a681817344e8c95026eb5b27d538a : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196d950a64014a99b4c135bed2dbc429 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/6127e6b5880c466d8f49ebfa530fcda4 : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e05e41bf8c08496ebc126e5c3291cc62 : SUCCESS in 1h 06m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/91a416e5451e4b64bc4a0bf0f580da1d : FAILURE in 5m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e27f494c5c1495f9447ff6ecbde238f : SUCCESS in 6m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3bdd4f6a6104f6b80a197613ea61bb0 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac285611b7814c6a937cadc62952540c : SUCCESS in 6m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f51b20370b284471ae416932584c257d : SUCCESS in 7m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91f63ee35e9c4bd9a26b6d4928bf1887 : SUCCESS in 35m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a092b1b6cc224cd299bbcb50f213c172 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/517eec7d94c24097bd8fd77725dae8e5 : SUCCESS in 1h 17m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2c344fa85a0e43fa8477bb47c304c7b8 : SUCCESS in 1h 25m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c982ac679334f308cd50c64405fefcb : SUCCESS in 1h 26m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29e3adfda9f2426a9ff2a66f0e06b9ec : SUCCESS in 1h 26m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/905a9a879e4f4b2796bd3b83fc2ae665 : SUCCESS in 1h 10m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/17d9eef978a94362a6ecfaff34af587b : SUCCESS in 42m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/923ac3e2d8f2466db48f353fd6ac4518 : SUCCESS in 1h 19m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f33df347f7094f35ac041c7870d463f2 : SUCCESS in 2h 09m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fc892b447074f39a8a4d0010118cfe4 : SUCCESS in 48m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f31a8ccb1ca4689818747cee8f0856f : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":3},{"id":"b22aa948ec6556c3b6b189387924ccb614c723ef","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-09-22 15:31:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_PowerStore_iSCSI/32ecb81/DellEMC_PowerStore_iSCSI/657 : FAILURE in 34m 06s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_PowerStore_FC/87fce93/DellEMC_PowerStore_FC/701 : FAILURE in 57m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"bdb092cfffee912b96261d9b8eb5301bc13f175d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-09-22 15:31:46.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/3/check/tempest-dsvm-full-focal-py3/f20a8f5/ : TIMED_OUT in 3h 13m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/810000/3/check/devstack-plugin-ceph-dsvm-focal-py3/6e6188a/ : SUCCESS in 2h 45m 14s","accounts_in_message":[],"_revision_number":3},{"id":"57de40aab25961525156d586eec5d801292827ba","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-09-22 15:42:16.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_XtremIO_FC/f459deb/DellEMC_XtremIO_FC/1042 : SUCCESS in 1h 14m 10s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_XtremIO_iSCSI/0d2de4c/DellEMC_XtremIO_iSCSI/928 : FAILURE in 2h 40m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"7a054e7587b102d98c2f86d05fa27b1c0def2c62","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-09-22 16:01:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_SC_FC/a28f09d/DellEMC_SC_FC/597 : SUCCESS in 47m 14s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_SC_iSCSI/b53a654/DellEMC_SC_iSCSI/565 : SUCCESS in 1h 27m 57s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"1511c06983c637e90ac3c64006363a7489526ed1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-22 16:09:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/810000/3/upstream-check/cinder-cDOT-iSCSI/c566475/ : SUCCESS in 34m 50s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/810000/3/upstream-check/cinder-cDOT-NFS/6550e71/ : SUCCESS in 33m 51s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/810000/3/upstream-check/cinder-cDOT-FCP/690ef17/ : SUCCESS in 55m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"1fcac77b970a10a14d3a042cb259ddc4ae523cbb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-09-22 16:17:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/810000/3/upstream-solidfire-check/cinder-solidfire-iSCSI/164ed74/ : FAILURE in 1h 42m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"bf1caa011c47a8427cc45568b926e7a67869c516","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-09-22 16:28:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/810000-3/PVME-iSCSI : SUCCESS in 43:35.30\n- PVME-FC http://os-logs.tristero.net/dell-ci/810000-3/PVME-FC : SUCCESS in 42:47.93","accounts_in_message":[],"_revision_number":3},{"id":"e7f2d05f42ee2175cba765eee322b53f37526c78","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-22 16:34:29.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/810000/3/check/DellEMC_PowerFlex/a431e37/DellEMC_PowerFlex/685 : FAILURE in 1h 39m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"892883afe24085f1b5b33a8966a2bd9db2daac26","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-23 13:44:48.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM.  Thanks, Stephen.","accounts_in_message":[],"_revision_number":3},{"id":"ddbd2c95c4a31d91deca9513966f13ce16c42bfe","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-09-23 14:18:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks like a good thing to get fixed.","accounts_in_message":[],"_revision_number":3},{"id":"ba4673e2c2929b00d55ae2abd1761efa7a6a972c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 14:22:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d177b934833824dca5675f6fcad4c7a98c7c78e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:17:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed31164a2b34b6781128f82045e8b09 : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0915c56bd10d4663babb33f3d4171115 : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfc1ef1bfe7e401084d50203ce942e9b : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed98028152044ec7a75526dd97d130f8 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/46f821d91bf94c2197c335b88846e9fd : SUCCESS in 1h 06m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8cc74e1321084bd7bcbf211baa15fb8c : SUCCESS in 1h 11m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef817b022a2c4a02b3b3071e8228cdaa : SUCCESS in 1h 12m 25s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/78a14668c54c46d6bffb481d5d9d5221 : SUCCESS in 1h 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04b7dc22da33488fb254fdae81237e04 : SUCCESS in 57m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67ca7d8b27514a5ab24dff5b5d81594c : SUCCESS in 48m 20s","accounts_in_message":[],"_revision_number":3},{"id":"c499a4d05a6a2cdbb6d4ef7f5b6a56789837852b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:17:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4ca9e329eabd91254d0220c8bb0b30ec1b5fbde4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:31:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2eb4c8ca954af6be77879cf2af3ef2 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7187dbfb33008441cd290c4a6b2b89c5d3dd541e","revisions":{"ce5bb8ebb1946529f41e53fed8c8fcd16157d1a3":{"kind":"REWORK","_number":1,"created":"2021-09-20 11:26:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/00/810000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/810000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/810000/1"}}},"commit":{"parents":[{"commit":"4c2f2b460baf69b0d0a02014aeb2fded59c19484","subject":"Merge \"PowerMax Driver - Port load balancing fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c2f2b460baf69b0d0a02014aeb2fded59c19484"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-20 11:11:34.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-20 11:24:07.000000000","tz":60},"subject":"Fix cinder-manage db version","message":"Fix cinder-manage db version\n\nWhen cinder switched its DB backend to alembic, the aforementioned\ncommand was modified to support querying the DB revision via alembic.\nHowever, there are two mistakes in the current implementation. Firstly,\nthe code called \u0027alembic_api.current\u0027 [1] with an \u0027Engine\u0027 object, but\nthat API expects a Config object instead. Secondly, the same code\nexpected that this call returns the revision, but that call just prints\nthe revision to the standard output instead.\n\nCorrect both issues by switching the API used from calling the alembic\ncommand API, which is mostly created for CLI consumption, to a\n\u0027MigrationContext.get_current_revision\u0027 API that is intended to be used\nas a Python API.\n\nThis fix mirrors a similar change made to nova as part of change\nI9fa7c03310c5bdb82e9a9c39727edb12eeae77f0, which fixed bug #1943436 in\nnova.\n\n[1] https://alembic.sqlalchemy.org/en/latest/api/commands.html#alembic.command.current\n\nChange-Id: I34f8d50cab626526cfd2416f605cc0f2a9325e88\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #1944118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce5bb8ebb1946529f41e53fed8c8fcd16157d1a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce5bb8ebb1946529f41e53fed8c8fcd16157d1a3"}]},"branch":"refs/heads/master"},"803319a4c25c61ffe7ba79aa120ec58d1c548dd6":{"kind":"REWORK","_number":2,"created":"2021-09-21 16:20:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/00/810000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/810000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/810000/2"}}},"commit":{"parents":[{"commit":"ed4d5585e4ef756977afb9388a4ce8c3b203d1a5","subject":"fixtures: Don\u0027t persist state in the Database fixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed4d5585e4ef756977afb9388a4ce8c3b203d1a5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-20 11:11:34.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-21 16:20:21.000000000","tz":60},"subject":"Fix cinder-manage db version","message":"Fix cinder-manage db version\n\nWhen cinder switched its DB backend to alembic, the aforementioned\ncommand was modified to support querying the DB revision via alembic.\nHowever, there are two mistakes in the current implementation. Firstly,\nthe code called \u0027alembic_api.current\u0027 [1] with an \u0027Engine\u0027 object, but\nthat API expects a Config object instead. Secondly, the same code\nexpected that this call returns the revision, but that call just prints\nthe revision to the standard output instead.\n\nCorrect both issues by switching the API used from calling the alembic\ncommand API, which is mostly created for CLI consumption, to a\n\u0027MigrationContext.get_current_revision\u0027 API that is intended to be used\nas a Python API.\n\nThis fix mirrors a similar change made to nova as part of change\nI9fa7c03310c5bdb82e9a9c39727edb12eeae77f0, which fixed bug #1943436 in\nnova.\n\n[1] https://alembic.sqlalchemy.org/en/latest/api/commands.html#alembic.command.current\n\nChange-Id: I34f8d50cab626526cfd2416f605cc0f2a9325e88\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #1944118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/803319a4c25c61ffe7ba79aa120ec58d1c548dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/803319a4c25c61ffe7ba79aa120ec58d1c548dd6"}]},"branch":"refs/heads/master"},"7187dbfb33008441cd290c4a6b2b89c5d3dd541e":{"kind":"REWORK","_number":3,"created":"2021-09-21 22:11:02.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/00/810000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/810000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/810000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/810000/3"}}},"commit":{"parents":[{"commit":"ed4d5585e4ef756977afb9388a4ce8c3b203d1a5","subject":"fixtures: Don\u0027t persist state in the Database fixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed4d5585e4ef756977afb9388a4ce8c3b203d1a5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-20 11:11:34.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-21 22:10:44.000000000","tz":0},"subject":"Fix cinder-manage db version","message":"Fix cinder-manage db version\n\nWhen cinder switched its DB backend to alembic, the aforementioned\ncommand was modified to support querying the DB revision via alembic.\nHowever, there are two mistakes in the current implementation. Firstly,\nthe code called \u0027alembic_api.current\u0027 [1] with an \u0027Engine\u0027 object, but\nthat API expects a Config object instead. Secondly, the same code\nexpected that this call returns the revision, but that call just prints\nthe revision to the standard output instead.\n\nCorrect both issues by switching the API used from calling the alembic\ncommand API, which is mostly created for CLI consumption, to a\n\u0027MigrationContext.get_current_revision\u0027 API that is intended to be used\nas a Python API.\n\nThis fix mirrors a similar change made to nova as part of change\nI9fa7c03310c5bdb82e9a9c39727edb12eeae77f0, which fixed bug #1943436 in\nnova.\n\n[1] https://alembic.sqlalchemy.org/en/latest/api/commands.html#alembic.command.current\n\nChange-Id: I34f8d50cab626526cfd2416f605cc0f2a9325e88\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #1944118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7187dbfb33008441cd290c4a6b2b89c5d3dd541e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7187dbfb33008441cd290c4a6b2b89c5d3dd541e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
