)]}'
{"id":"openstack%2Fcinder~810737","triplet_id":"openstack%2Fcinder~stable%2Fxena~I6acb96d8951a85a74c1a33d100cd464837de214a","project":"openstack/cinder","branch":"stable/xena","topic":"drop-sqlalchemy-migrate-stable/xena","hashtags":[],"change_id":"I6acb96d8951a85a74c1a33d100cd464837de214a","subject":"fixtures: Don\u0027t persist state in the Database fixture","status":"MERGED","created":"2021-09-23 18:24:47.000000000","updated":"2021-09-24 18:32:37.000000000","submitted":"2021-09-24 18:29:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810737-drop-sqlalchemy-migrate-stable/xena","cherry_pick_of_change":810288,"cherry_pick_of_patch_set":1,"meta_rev_id":"d468b8b8fac2ef46824a414a299da07986745da2","_number":810737,"virtual_id_number":810737,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-24 18:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-09-24 13:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-24 17:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-24 17:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-09-24 13:27:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 18:24:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-09-23 18:46:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 19:36:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 22:35:19.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-09-24 17:20:29.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"235b0d0ffc27ae54aa21c89e579fc26cf11ec6e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-23 18:24:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b2299df8587fa4f7e5114efc1b87c486b0f459a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:46:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b9659320a343fda107be68386884da : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13e229a1be94427cbfb5d8a2a5bb1c3a : SUCCESS in 11m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fca8321421d74cf49881062778b5e0b9 : SUCCESS in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aab4304375214497b4385e72bfb5dd0d : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2d499389f947ed8439c109c93bee83 : SUCCESS in 12m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8e331bc81f1e49ad947cfe7dafaa8f0b : SUCCESS in 15m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/543c5b04e6604785a1926d2d96e7a76c : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c598e31ebabb69a560429a65cab2c52ee003867","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:49:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd8bda1274f64045ab24c0bff05e1c9f : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d09025b42c146e4a86b4b853a44612b : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c74f5826092c9191affdace9dd651f44af747e47","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-23 19:36:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/37/810737/1/check/iscsi-driver-alletra-6k/425c04f/425c04f.html : FAILURE in 1h 06m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8f7b652821bd0a3ce483f8095b3a4c68319be731","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-23 19:40:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/37/810737/1/check/iscsi-driver-alletra9000/4794101/4794101.html : FAILURE in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"aeccaa471a7eafc3f6f261033275e7b5763f1629","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-23 22:17:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/37/810737/1/check/iscsi-driver-primera/466e15f/466e15f.html : FAILURE in 50m 13s","accounts_in_message":[],"_revision_number":1},{"id":"76ce7fe4e8667cbe3d86683a781f68cbc0635e8f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-23 22:35:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/810737/1/check/dsvm-tempest-cisco-zonemanager-job/104e553/ : POST_FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5cd715a0bd49a25c8244dbc7875dc9b17845bbb1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-24 03:01:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/37/810737/1/check/iscsi-driver-nimble/4fd7378/4fd7378.html : FAILURE in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7cbc9202b989163d08f0d77ca8bc2b60f3ae7b7d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-24 13:27:24.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nWe need this in RC-2.","accounts_in_message":[],"_revision_number":1},{"id":"e9f1a197b4d603e83119088516f623267e9a3bf1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-09-24 17:20:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nWe have two cores on this.  No reason to hold it further.","accounts_in_message":[],"_revision_number":1},{"id":"33572f679806cee66db522b698c4871569b850ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 17:22:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ded9ad217963f00f5e16938f1b606addacd729b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 18:29:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c00c26242254ef1a51b78bfe71c0971 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958e205ea1be4aa9b2f397b67e07fb18 : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d67d2984c15340d1aec862fbbd0bc39d : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f5e6c2a7c84c45928b5f5b3c711783 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"16f1b6bffc5d40baa4a4291d1aff0be3cd82cc0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 18:29:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d468b8b8fac2ef46824a414a299da07986745da2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 18:32:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a604450b30842468b6ea6b8bfe96a6a : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1fb66278aca8aab4243dcc22a9eb89807c16f1c","revisions":{"e1fb66278aca8aab4243dcc22a9eb89807c16f1c":{"kind":"REWORK","_number":1,"created":"2021-09-23 18:24:47.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/37/810737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/810737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/810737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/810737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/810737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/810737/1"}}},"commit":{"parents":[{"commit":"e45d81f37c27ce3c954bb50ee1947ecd9f8b41ea","subject":"Merge \"RBD: Call trash operation when plain deletion fails\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e45d81f37c27ce3c954bb50ee1947ecd9f8b41ea"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-21 15:22:53.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-09-23 18:24:47.000000000","tz":0},"subject":"fixtures: Don\u0027t persist state in the Database fixture","message":"fixtures: Don\u0027t persist state in the Database fixture\n\nThe way that the Database fixture is implemented appears to be storing\nstate. We\u0027re configuring the database engine as part of the \u0027__init__\u0027\nof the fixture and then caching this created fixture. I suspect this may\nhave been a performance optimization or similar but it\u0027s wrong and is\ncausing issues fixing bug #1944118 since that fix ends up with us\nattempting to apply a database schema twice. Avoid this by avoiding the\nglobal state entirely.\n\nChange-Id: I6acb96d8951a85a74c1a33d100cd464837de214a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nRelated-Bug: #1944118\n(cherry picked from commit ed4d5585e4ef756977afb9388a4ce8c3b203d1a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1fb66278aca8aab4243dcc22a9eb89807c16f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1fb66278aca8aab4243dcc22a9eb89807c16f1c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
